)]}'
{"id":"openstack%2Fcinder~966238","triplet_id":"openstack%2Fcinder~master~Iac00cddb601655f0adde7cfef9902bedb882f587","project":"openstack/cinder","branch":"master","attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-11-13 09:04:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iac00cddb601655f0adde7cfef9902bedb882f587","subject":"[DNM] Skip virtual size check for Image-Volumes","status":"NEW","created":"2025-11-06 06:42:43.000000000","updated":"2025-11-13 11:08:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3b4bf4a4e6802c338ab2c9342b1ab08e0d31632","_number":966238,"virtual_id_number":966238,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-13 09:04:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 06:43:48.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-06 07:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-06 07:39:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-06 08:47:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-06 08:48:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-06 09:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-06 12:34:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"5f039f0ee777e7f6a721288ca32f89fe322edfc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-11-06 06:42:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27127ab951f52ec05bf69fef0531efe9d5d87d7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-06 06:43:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"96eadbac1b0fdcaa5c32ef65fa16407dee272d8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 07:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4258c7f97d4986ab0bef896d8a5047\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f74c0834232c4b409ee3f0d96a038e6e : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77b783a25bea4e9383c08ab3dd167071 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/587dec78b49940e19d4cc406e774eca6 : SUCCESS in 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1272871ba99ffe26076d3efc84286b448345671c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-06 07:39:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iac00cddb601655f0adde7cfef9902bedb882f587 .","accounts_in_message":[],"_revision_number":1},{"id":"21f0123821a0f786a49c015186a5d91efd5b6d4a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-06 08:47:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966238 : FAILURE in 36m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e59dbd7e414125007ed8e21cda239013a6cbb924","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-06 08:48:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2c9693bbc8f04d3b8bed54cdc5a79f55 : SUCCESS in 45m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/648efe4370ca4c488765b3e456cc89b4 : FAILURE in 53m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3f1f75d9bd5af44d43388138c37c95399c33f53f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 09:52:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/277fa81cda6b4f0aaa9ab45bc3474e97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d548227eb44e24aa8ec82416c845dc : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8c02f6823ee4ac4880178b9b2a0b1ac : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79f7748ca92f4a0bab3fc26245e376ed : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab4b7006f134fd9b4a1318f60d82e87 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/774821fcd7044b4d9ab067f16b2d8475 : SUCCESS in 58m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b36c2fd6f534867a4061dee5698a370 : SUCCESS in 45m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/443cd65f65904747acd8969b06436f24 : SUCCESS in 1h 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a357671d84d4c74bf546064bad3e50c : SUCCESS in 35m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac7ccf38366f4a31aac21f05a9b6c8c4 : TIMED_OUT in 40m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1daf52f6eef041e38d0a004e14467586 : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b2cda7cdfd674e48862ef96843c96332 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b8a21dfd36194603a0b7faf72f6892ff : SUCCESS in 5m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5f5ffcd7f26c40c28a095041e96b2aed : SUCCESS in 8m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e0d4cb1e6b847a6b3645d4a3d9c1be2 : FAILURE in 41m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c4e8a52569044d00ae29778755e9b8aa : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/555643ff7d4947c99ccd60a4c2271e87 : SUCCESS in 1h 53m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6243d3ddd9ec412ea1cb5db309d4b71e : FAILURE in 2h 51m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fdb343b3fe3f4987bec3875133b30e79 : SUCCESS in 1h 49m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5f06ab937eec4c12b9e105d78cad6e62 : SUCCESS in 1h 28m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a13b82355f904dd493fde93dbd05eeb6 : SUCCESS in 34m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/681c4476df954564b807f708d1130d9d : SUCCESS in 1h 07m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6086993d760f4b7ab9aa7e85b3157b3c : SUCCESS in 1h 44m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e8481a1a64645a1bb2012002f49c06f : SUCCESS in 48m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/55b88f8b679e406c8ab7d38d05352c51 : SUCCESS in 1h 35m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/43e60ff9fbb34d44b916e4be2b3fb39e : POST_FAILURE in 3h 05m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/258e7e9654be48f9900e881938f23890 : SUCCESS in 1h 26m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28707c2cad2d43c5a48f500b8b8e89a3 : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d57b3a08be3348766a82cc8c53bbaabbaaab2872","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-06 12:34:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2325/.","accounts_in_message":[],"_revision_number":1},{"id":"d827c7ffa6ed1499e60ad250b2f5c10e075afa34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-11-13 05:43:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd6a78f9ee91f766e9c1a7bb0e5e9f75094f5209","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 05:44:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"de06d9f86db5a73cd90931390979644fa92321be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 06:23:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b91806fdfba44d59df6be50d5c15850\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1d00d806a0c44cda790aa6c34d0af02 : FAILURE in 29m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2cc07cee3fd490dadc27c7c70999628 : FAILURE in 31m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/239a0689bab34871a4ab61ec2ba251ca : SUCCESS in 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e10c1a76a5cda9d96674947ddd897c19aa60e32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 06:39:14.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iac00cddb601655f0adde7cfef9902bedb882f587 .","accounts_in_message":[],"_revision_number":2},{"id":"63da0f9aa8ec7364c105c6b539ef5140223379a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 06:55:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2432/.","accounts_in_message":[],"_revision_number":2},{"id":"7cece6568de9c87737b68c759c176ec679854800","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 07:03:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2546/.","accounts_in_message":[],"_revision_number":2},{"id":"6f7a049f2c3fec0f67f1a9480a13f20bc57a43bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 07:03:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bf0bcc6233db442abb35388c44f3e1a3 : FAILURE in 14m 36s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2f136cd7695848caa4fc3aeec337d066 : FAILURE in 14m 22s","accounts_in_message":[],"_revision_number":2},{"id":"08d4254d88faabc7c0990e0505752c687aebb6c7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 07:16:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966238 : FAILURE in 36m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7e9e107a6ad637795dd3d65835347b7063781dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 09:04:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99df2ff6a28b4fef8af0f6776651e646\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8459e874b5b84d8782baf9ed666acb6e : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fee136f7c46649aa9f7ce0a4e1fbbcf7 : SUCCESS in 8m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ef5f7bb322e425abfa64ee02a553d22 : SUCCESS in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e60500a0b6448c963cb1117abb2178 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/6707218f61ea4aaca3da3d8e62544794 : SUCCESS in 58m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2652e8a31e17449492d409f7a0ba4163 : FAILURE in 5m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b97ba2d0bd94d4389c23dd7f958f8a0 : FAILURE in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4189114d64349399936440cd05176b1 : SUCCESS in 58m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cee74e8f8d5e4c7b84c66d70d4aa595c : SUCCESS in 19m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f762806e95d4736b6a9c4c1aa2e01fb : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3cfeeba70b174b6cac37acf6524e98da : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff8f8adab92549a18ce44e79e91fe59f : RETRY_LIMIT in 1m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aa0d59df49bf4320850aecb025845d39 : SUCCESS in 8m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b4d6f4f6a0943b2839e72621f7902ca : FAILURE in 5m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c33d71c4ff8c4fd3ad9a57c9f835b502 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aecdaa34d4554bb4817a986c127a8041 : SUCCESS in 1h 50m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c90ff6b8a8ab4e5292dc180a31be4bfa : FAILURE in 2h 03m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a68abe732b14e0da6f24f6505559a64 : FAILURE in 5m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9f47c00fded6497ebe1fe377582e74d2 : SUCCESS in 2h 19m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/838003e2e374410f9a93ed1bbaef014c : SUCCESS in 19m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc25847d72054d6e8d04eca22a78d8bd : SUCCESS in 50m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/328e4affdfea4ac1b2e79f0c1e9703f3 : SUCCESS in 1h 39m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/267ed0600b154ca99df9ce828db10b5a : FAILURE in 5m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/611c136595234a4bb19ed50af16fab2b : FAILURE in 5m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/70be730bd9b84a988fcce384b6752b84 : TIMED_OUT in 3h 11m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffd279e1512c4eee9bef79a1bd68e973 : SUCCESS in 1h 45m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2042a899cbd145ea87f0e0e42c7372fe : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a3b4bf4a4e6802c338ab2c9342b1ab08e0d31632","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 11:08:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/18da94821fe44a1dbe10f28f73eb5629 : SUCCESS in 43m 44s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f61d19cb875d461b9a3dd32a7fd1b19a : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd52a51cc5c3a7b85fdc5c82146da749b674f95b","revisions":{"24d4d3ec8d201ff25c83f60f90452b9d44d55d2a":{"kind":"REWORK","_number":1,"created":"2025-11-06 06:42:43.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/38/966238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/966238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/966238/1"}}},"commit":{"parents":[{"commit":"a04982692a818c23d957887bfd45903f002cd5fe","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a04982692a818c23d957887bfd45903f002cd5fe"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-06 06:41:39.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-06 06:41:39.000000000","tz":-180},"subject":"[WIP] Skip virtual size check for Image-Volumes","message":"[WIP] Skip virtual size check for Image-Volumes\n\nChange-Id: Iac00cddb601655f0adde7cfef9902bedb882f587\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24d4d3ec8d201ff25c83f60f90452b9d44d55d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24d4d3ec8d201ff25c83f60f90452b9d44d55d2a"}]},"branch":"refs/heads/master"},"bd52a51cc5c3a7b85fdc5c82146da749b674f95b":{"kind":"REWORK","_number":2,"created":"2025-11-13 05:43:28.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/38/966238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/966238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/966238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/966238/2"}}},"commit":{"parents":[{"commit":"4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-06 06:41:39.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-13 05:43:01.000000000","tz":-180},"subject":"[DNM] Skip virtual size check for Image-Volumes","message":"[DNM] Skip virtual size check for Image-Volumes\n\nChange-Id: Iac00cddb601655f0adde7cfef9902bedb882f587\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd52a51cc5c3a7b85fdc5c82146da749b674f95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd52a51cc5c3a7b85fdc5c82146da749b674f95b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
