)]}'
{"id":"openstack%2Fcinder~966503","triplet_id":"openstack%2Fcinder~master~I1a04ab75cdeb7935a5799960344efaec248c4979","project":"openstack/cinder","branch":"master","topic":"ci-test","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-28 02:16:23.000000000","reason":"Change was submitted"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-02-14 10:40:24.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"37940":{"account":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"last_update":"2026-02-28 02:16:23.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-02-28 02:16:23.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-02-28 02:16:23.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-02-28 02:16:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a04ab75cdeb7935a5799960344efaec248c4979","subject":"Hitachi: Use override_config in tests","status":"MERGED","created":"2025-11-09 00:19:06.000000000","updated":"2026-03-01 07:51:31.000000000","submitted":"2026-02-28 02:16:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966503-ci-test","meta_rev_id":"bd00017bffd476899d3e859a08c167f5c1fd496a","_number":966503,"virtual_id_number":966503,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 02:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-02-27 17:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-02-27 17:53:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-02-27 18:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-09 00:20:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-09 00:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-09 00:57:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-09 01:17:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-09 01:21:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-09 01:21:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-09 01:29:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-09 04:17:40.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-02 16:33:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-27 23:38:10.000000000","updated_by":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-01-27 23:38:10.000000000","updated_by":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"reviewer":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"state":"REVIEWER"},{"updated":"2026-01-27 23:38:29.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-01-28 00:34:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-28 01:01:50.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-28 01:11:51.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-29 19:38:13.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-02-04 15:23:42.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2026-02-04 21:22:35.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-13 18:48:05.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2026-02-27 17:57:36.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-27 18:28:29.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-27 18:32:33.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-27 18:36:21.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-27 18:53:20.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-03-01 07:51:31.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"c51a9435db1632d968672bb34c6d513cae6a4ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-09 00:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94dcda6a02ee693c7bd2a81e1996b6d4fc73d419","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-09 00:20:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"50fca7d50c014368bd9d4bcfcb3dde4bfc5645e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-09 00:43:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3dd8d86408488b910827e75f93f335\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24f6dd5370464dcfa93f146ded2e0e3b : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/088eb426b19b4375ad2bebdf2cae0206 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22bc256321d04f2bb5ef7d100b08d9fc : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1ac9ecbba36cc431ad4bb33d1fadf37737749d0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-09 00:57:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"08f8240f1c71d7b2f8ac861930226a297808ff84","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-09 01:17:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":1},{"id":"db9bbbbd4ac493e3cfe994562113f7f83143b2bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-09 01:21:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2348/.","accounts_in_message":[],"_revision_number":1},{"id":"fe51ee39bdf065ce5ebcfd694e8784dd9cab0b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-09 01:21:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a89b7edf2814048839cb1ff264e979d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3781ca036df4fbcace2ddd3e97b2938 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37bdfe35eaaf43848fa367bb451fdc4a : SUCCESS in 15m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8a94906b7094a87a3a491eb505f2c50 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cd3eeebb994f4599945024fc6a5f79 : SUCCESS in 13m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c5b5cc496ff4cafbd8a26d9b05ce594 : SUCCESS in 57m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d6ec5a625f74c29bbb137254eff7e24 : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/49f6fb5715ac482cb15e8aff7fb849c5 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3dbc2fdd9944ab03406d42974dcafb378fa8b942","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-09 01:22:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2462/.","accounts_in_message":[],"_revision_number":1},{"id":"759503a373697254340e63a2c874bd0b3fda3e25","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-09 01:29:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ba942dd30b064b328d1bc62837dcc399 : SUCCESS in 44m 45s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/62b113e3f4e9461fb457ecff15a78fef : FAILURE in 1h 06m 55s","accounts_in_message":[],"_revision_number":1},{"id":"681cf971b721fe2cde7e5af52ada25f9b63a0087","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-09 04:17:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-1-fc : SUCCESS in 59m 33s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-1-iscsi : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2be2dd73ef06cfd4a9ae340333074366489c6967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 06:11:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_27615\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"_revision_number":2},{"id":"7c22235ec7e15379a90b8ef5222842058f6aa5b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 06:12:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b10c76f727832d4834408aca9881c0de371e1405","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 06:38:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c72d41309248c4897de16ea2f4e0d5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de12e80d6d3b4e96891441d845916adb : FAILURE in 24m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b6821feea314b76a3e669f8c32aa81f : FAILURE in 24m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ec740e1405a44859f1db87a865e59f2 : FAILURE in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca8de2159c00ab0690b9cc743d962a9e3285d291","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 07:07:07.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":2},{"id":"1b892bdbbf214976b73ebcb4b978541290c6f1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 07:13:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1de954498ea44b8989b9fd7bb022d73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2750fd59df5742b68907b27f1f0f317d : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26a6fa443e6646b28944dec1b7b6d1cb : FAILURE in 11m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16a47ca2690c452f9843d371918045ab : FAILURE in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862226e06595493b9e5ac37a02745bb2 : SUCCESS in 13m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25d085b3a8744795ad53962cfd1e3664 : SUCCESS in 57m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/affcfeb0e2154b3eb389fec41cbb5700 : FAILURE in 15m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9330262e295f4fb68f67b43357cb9874 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"147821dcbe60d7854ecfebf32f89d9ac975e3890","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 09:54:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2601/.","accounts_in_message":[],"_revision_number":2},{"id":"afa0499a3318e6b70ee6be95b7facc92abb5fdce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 10:04:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_27615\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"_revision_number":3},{"id":"09ff77aa81ebf8d867775de912f4e70092398527","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 10:05:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"227ee4155971db825703e0028f86a7aeb36759b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 10:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2487/.","accounts_in_message":[],"_revision_number":2},{"id":"d8e3ec594b9385f62e35c2ee96279bc07fbfb1e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 10:34:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983189aa8302413388f89c429a11fcb0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/310e45dc17454bfaa3b4d528aed18cf8 : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82ca53e7cbb14de0a66590a496c51139 : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/04fc8dd78f5c46c7aae8ce625166a23f : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"50534aa09005c08ca4b0cdff7967b40ee667a659","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 10:59:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":3},{"id":"a5852672cdbfd3639150cc8612b40cb1a74d89ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 11:16:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0152e7212054ea1b793fda1229bafaf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088d690825734ec5b189d07c1905650b : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e76ec1fa346453eabc5af65e6adbae4 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e32768a6faf7448bb50b2247e507dd50 : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63795b4aa3d147a9845893d88f8e32c8 : SUCCESS in 15m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23c54a6e5f3d414aac277901488f1dd4 : SUCCESS in 1h 03m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a04b9dd1b1f45d4aede44892ef8b992 : SUCCESS in 19m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6402545fea49460e94ddc6edf896a465 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"64fb1913c1d981857bba41f61febe75e2e32901c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-14 12:09:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b78037f9855945d8ae31ae20d6a25ad7 : SUCCESS in 47m 51s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6ee38874350f4faa97b299d4e8bc716f : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b3c89de3aaade608de29f25540f974ee1fb81593","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 13:08:21.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2604/.","accounts_in_message":[],"_revision_number":3},{"id":"7810d74100a3190c3a98b2ac2659617a22609c5f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 13:18:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2490/.","accounts_in_message":[],"_revision_number":3},{"id":"8e23db90fcfe47499585140fcad8fbc847bdcf65","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-14 13:52:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 1h 12m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fcf2177e6f9acc00b88680a79f65280f57cec674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-15 13:59:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_27615\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"_revision_number":4},{"id":"a0f32838ffce864c0dc5fae5c9a01c2c1e9f1981","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-15 14:00:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"7b19ceb1fbcf18acf85cee740bda34c12c542626","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-15 14:03:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5cee2675afab4e629e57870b35c3c7b6 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8a94311aecbe4a9bb65ef5c8b3775350 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"7657db176325ae5351b12574bfbb4ab4467a9c12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-15 14:26:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc0490afc91425e85aec730eaa8972f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c7216e505fa496e86ddb7e464993f33 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37d50565674546c18fbf17e9a8a8af89 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0144b18ce1844405a63beb9e672bc7bb : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1ef920124e36d8ddf2bbe5ac63494f5853d1db02","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-15 14:56:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":4},{"id":"35f17a2ab8141ca2ada9dcaf13a2221208e3ce27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-15 15:03:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21b93e31d42d48468db8bffc31612b4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a012af16c84fdb9feaa4ac91d011ab : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff11b3c916fc47ca92bb3ddcce6504ff : SUCCESS in 10m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0672e71bea2e469aac2be8ecbf6c4960 : SUCCESS in 12m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19af988b23064fb28900c142632f7679 : SUCCESS in 14m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0751ba5234f94ac69b4187aa09b52a5a : SUCCESS in 58m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/264f243a545847e6a482fdb56f9f967c : SUCCESS in 24m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a4069ac65464c91b986f16a18d0e22c : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6301ecfb913a56c6955fd1d3e35edc9c8e3b9d3f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-15 15:52:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 1h 10m 51s","accounts_in_message":[],"_revision_number":4},{"id":"3efc57766aa5fe651aa947093a106176dcd1a454","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-17 16:21:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/094f9fa9430f4519bf411391fc1fe054 : SUCCESS in 44m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e6add346a690437385f20a99cf692462 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":4},{"id":"87f7208f829716a25309770a8bbc45ab81bc8ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-18 11:11:14.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2fa1e1bd746bf687875edae756aade2e98cace1e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-18 11:12:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"e341214e3eccbd986bfd5ed3326b1aa10334b816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:43:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1451073f508344d2bdfb28ab2cb103a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0920afa85b7443deb9053b0459327a94 : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/518ac3572a0d4ff9a8460f85b72bd308 : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/184c20a2fac54e8b8e08185b936db6c5 : SUCCESS in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65aa76e123de9a3b1cee87376475ebc9a214a69e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-18 12:08:54.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":5},{"id":"73f5ba1423bece7f0e1c9f34564df97d24b1b906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:13:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b624e54511f45b58f187b359f8ac851\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094bae6713f64a1c9cd45993b9447ded : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8c83291364f4b82b526baefd5afc7bc : SUCCESS in 9m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ffdc153a98c444e89e7c95bef9d138e : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dec890bfb0f4dc58212f41019a3a381 : SUCCESS in 13m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/697f5a886eed402782c098f66bf8625e : SUCCESS in 57m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d08f04e9aba34a9ba28ef3b0e0e7e705 : TIMED_OUT in 41m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/45512d8fa41c4d10a2d7c14e1c5c0e22 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":5},{"id":"229d2092541d5a809b8a6559fc30be21f8c8b91f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-18 13:07:58.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/22559e6ba48844a186b364739f391b30 : SUCCESS in 49m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bd9b91d741874777ba804bd76d291232 : RETRY_LIMIT in 8m 58s","accounts_in_message":[],"_revision_number":5},{"id":"d5fb9e534e0c743ea3d7313c17ac9a7c1f5001cb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-18 13:37:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 1h 11m 32s","accounts_in_message":[],"_revision_number":5},{"id":"648478b544fd3447241f62d5d9a148a1092d5e17","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-18 15:17:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-5-fc : SUCCESS in 1h 1m 18s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-5-iscsi : SUCCESS in 1h 8m 47s","accounts_in_message":[],"_revision_number":5},{"id":"73d77e0a94a4ed14d1fe60c6851a5a3e7648533c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:15:46.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2493/.","accounts_in_message":[],"_revision_number":4},{"id":"b190d754a44791fec16ea4d0f45daa3b704f7086","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:17:03.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2503/.","accounts_in_message":[],"_revision_number":5},{"id":"0636714ec25b1374647e681020c503b1d273c538","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-12-04 17:31:47.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfab46c5f15f628b8356e3f94b550605adf1bad5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:33:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"66b504f9c0b519bcee90530480b4280d305cacc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-01-27 23:38:10.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8ad8bfa2e0990a402a6d222f50c7ad88e62b4c5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-27 23:38:29.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/679/","accounts_in_message":[],"_revision_number":6},{"id":"70cab534675894926cf428ad02f2ce5e3ed19bf4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-01-27 23:38:30.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/966503/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/966503/6/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"8b3e787b4aece3382552db5537c09b7f76ed601b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 00:26:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad77a5f3bd694d24865737cc305fa6f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec1dc8bbcdb45279e501370c290429a : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68b8911137cd4330bfd19968304a8a91 : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b21b9222c6248ef90ea05c16d46a77e : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c65d25a257f42098c7883592acb9085 : SUCCESS in 7m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1c8920f06a343b9890268be31f39204 : SUCCESS in 37m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3388775c742a49bdb1f3590a65493586 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa252ed72e524d1cbf8cbf78215db8a7 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":6},{"id":"a611328f1e596c12071e45dc3bdffdf71075b979","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-28 00:34:17.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_01_2026_00_27_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"311688ce20d5e722854432c124d691cc77855253","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 00:35:24.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":6},{"id":"c4fd1574be3b5530be5a1882b2e38eea2a886a06","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-28 01:01:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/e351cf1355b64db6aa313702a09acfee : FAILURE in 38m 50s","accounts_in_message":[],"_revision_number":6},{"id":"2edb257210a6328569d4ef35429194e629898b6f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-28 01:03:16.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 40m 04s","accounts_in_message":[],"_revision_number":6},{"id":"59967b05b35d4e30d6873c1d477724ad1cc1c35e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-28 01:11:51.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_369/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":6},{"id":"79250d092056a97547890223c1d8d26c2ad573f8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-28 01:42:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/051039021a48457d8683832a45386de6 : SUCCESS in 44m 17s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/795cbee17dfc47f3abb77acd6a0dc509 : RETRY_LIMIT in 15m 16s","accounts_in_message":[],"_revision_number":6},{"id":"9c5b8a3b98891dcde30fd9cbe4faf9c4a1838919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 08:19:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d1eff429c084fc78b298970a231386c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81fc38a490d14371b3e30f118279e1cb : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9628520c02e47b4be6d784f1c353ba0 : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/11ec22410b7c4a318529b3bff7bd2cff : RETRY_LIMIT in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52099e0200dd84998e04237ae32d265940007e2a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-01-29 19:38:13.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"72227d5ab6c1ce7ccfd1ba9e0cba36b2db1d16b5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2026-02-04 15:23:42.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/03/966503/6/20260204200705/NEC-V-ISCSI : FAILURE in 3h 26m 35s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/03/966503/6/20260204200705/NEC-V-FC : FAILURE in 4h 16m 35s","accounts_in_message":[],"_revision_number":6},{"id":"cff637559ee9d1222c71e786ab1c310c323bd678","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-04 21:22:35.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/03/966503/6/20260204200703/NEC-ISCSI : FAILURE in 10h 11m 54s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/03/966503/6/20260204200703/NEC-FC : FAILURE in 10h 15m 29s","accounts_in_message":[],"_revision_number":6},{"id":"d6b616bb339b50a1dfe081fc57fbbb9c98a9bb5a","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-13 18:48:05.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"63af9ff1d689423054fba80111c209a0979a519f","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-14 10:40:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"09d23ba38114a51861f6e7469cb78e09efffdcc8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-02-16 03:30:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-6-fc : SUCCESS in 56m 57s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-6-iscsi : SUCCESS in 3h 22m 54s","accounts_in_message":[],"_revision_number":6},{"id":"e7cf7335eb2b6c9bff080211db4227c36a6b23de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"real_author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-27 17:53:56.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_37940\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"}],"_revision_number":7},{"id":"4f35fd8924b8240d02a60b9465cdc2ee06bb47ef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 17:54:15.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3082/.","accounts_in_message":[],"_revision_number":7},{"id":"f3c48594ee4ed52ffd221519d7e214e01f103cc3","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-27 17:57:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c596f3ed38d14233b3b000d3c86f3542 : RETRY_LIMIT in 12s","accounts_in_message":[],"_revision_number":7},{"id":"c2804be7a00e6c10f4edec37314c11803bcc773d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 17:59:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7509446a70f34c04b454526537a49d9a : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/237527c7df19447188e42a67ca18006a : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":7},{"id":"d09d44a3d3763705d069682ae8b804e222ad43c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 18:00:49.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3083/.","accounts_in_message":[],"_revision_number":7},{"id":"d4b4a3dc2ab8b2105c76f41847fcedd0e5cbcfb3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 18:07:36.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3249/.","accounts_in_message":[],"_revision_number":7},{"id":"eb4e055c6b1e18475cccd756582a4533e1f39191","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-27 18:28:29.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/85ae1b1efdcf4caf9d12dbade00b363f : FAILURE in 34m 03s","accounts_in_message":[],"_revision_number":7},{"id":"8c94fdf015b4079f506d55015fcb1d3f85b149d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:29:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f92921eb9d484b5ca557227dda31bfaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f11944e37640eb9cae96eca59f071a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcbb3b1a369f4a639ff897216a02ba9c : SUCCESS in 8m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc820c13af404e23b46b74c9555a419f : SUCCESS in 7m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/869aa57224c848f4b620ebd283f44d7d : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30a4dc26aea4d6e97b707c267a58f84 : SUCCESS in 8m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cd5d404bddd44d5ae9bd17032c26db2 : SUCCESS in 33m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/078d4a80abbf4d9d93330c38177dff60 : SUCCESS in 20m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a7e9db7997d24df988a94ed5a59c3173 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":7},{"id":"f49105a185e3d41aa19eb553dcb68da1e89bf1fe","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-27 18:32:33.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/966503 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"5ba8df4d5a4164455164fa7aa92c61e18ea02a99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-27 18:33:16.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1a04ab75cdeb7935a5799960344efaec248c4979 .","accounts_in_message":[],"_revision_number":7},{"id":"a8c51feba7e06e135c50b1dc3b79ca9d858d9537","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-27 18:33:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c2380c800d704fc19bd4645318202e58 : FAILURE in 38m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6d71858d28d2cf5942fe2edc683cf89154287691","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-27 18:35:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966503 : FAILURE in 39m 39s","accounts_in_message":[],"_revision_number":7},{"id":"ae271d8bd50a31d44eaffd8bd39bf62fb6fd604d","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-27 18:36:21.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/966503 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"218fb8412065c855f728f13bb10222529187050e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-27 18:53:20.000000000","message":"Patch Set 7: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"356228274640d7bdd367950627be613ffa015b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:53:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d01f02d6fe4a17a463471b6635fda05a7d2a2296","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 18:53:56.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3085/.","accounts_in_message":[],"_revision_number":7},{"id":"5d65f8ee7b822aa876c0e658bf591500262f8c6e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-27 19:22:54.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_880/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":7},{"id":"69c576dba0ce9c5ac74da91510fa0def4caae340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:53:08.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9a65cac1b0e42729c5dfa7235b1a476\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec110910a35849929620d148ac81fd26 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4ad4888bbff4ceba7c3c1b95dde54eb : FAILURE in 12m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86f37a6aeb444c01acd57a3f61959bd7 : FAILURE in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eeb0803be08480e8fa5d9675c8686fc : SUCCESS in 15m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c9d66f4d5664f2dabe48762c7038a26 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":7},{"id":"1a7240fd7f3b509d53e213d2806f4c36c601726f","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-27 20:10:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"73b397db1f97b072e8da720ed16e32f34e43f858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:43:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e12f689750483d9cb4214dc419a28d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad1103e4b51a41678984031dd07a5466 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/016bc6e7857046c6abed1a79023629d9 : RETRY_LIMIT in 5m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c3ccf880f1240b6a8fb56cfadb38535 : SUCCESS in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f26fe9d64ab5bf5418b515888bbe3fd34c9bcab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 21:00:24.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c84c89aa2d1c479396a0b420c4b6adf8 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b6df70df7b1d45b386c1294f1cdc8982 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":7},{"id":"3c317f4a916f0ca547dd9a5461aa6952c5d3d6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 21:15:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d243e37dad7d4643a804e5740808a5fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563b98297df847aa8dd754af7938a964 : SUCCESS in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b41d88ccc604d72bcf3601264041d97 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b324f7e7d114eaab751f17641a67c54 : SUCCESS in 10m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03d795516e824ee787c85b98427fa9c6 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e528cb17e2347c5821b512310e54ba9 : SUCCESS in 14m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b026526a81c34d698009ec436da3992e : SUCCESS in 56m 53s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/62b1002561cb419c81cb1bd4039bbdc5 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c9df0cf4b62d4ebca4366da43c7ad5ce : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":7},{"id":"ceb0529b363408100a9244fa0e22d97cb0b95817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 21:16:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2e18cbb28c3fb0cdbf737b31364993358d419113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:18:23.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ec9ec952b084d8895a162c10fb36fd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a47a45d3df49a1a5ce11893a60088c : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dd6a4a841ea4352bd90e3213a185342 : SUCCESS in 8m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/178d86b8cf6347a2bf3b7b8138e00c47 : FAILURE in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0da17a12094b04bd3898ed49b5e347 : SUCCESS in 14m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/673caa7ed88d41f48bdff9f6a054583a : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":7},{"id":"7800370182851a7fc9940e266c485ed4185d94b2","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-27 22:48:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5967d8e65f387d3927d9d205e39bca63e9fe9393","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 22:54:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/029d134f6a954188a75b78ad2ddeb70d : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/02402c5751954dd6a81534c94495239e : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":7},{"id":"e6696af87b00c509dd2ac01ce80a2b751664833b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:07:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35cbd751cb414460a502e8e819fc87a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c82d44b831294f1a86669f30bd9ea584 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72403ae06b7c4121af0ed27873f82532 : RETRY_LIMIT in 4m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/517c728383b24736bffacab4e2630848 : SUCCESS in 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4445eadce29bdc9d24165146b6c83329b9c6c102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:20:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f754135cd7a64375a4269c1aa55e4fd1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b1df2cf1319404487787a9e6e414f45 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16e946476cff4126ac561f3b8c1e40a1 : SUCCESS in 7m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/533d8ccea1344af8b00a699d555c3cd5 : FAILURE in 8m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa70cd1275b54ca18d1cde59f3b24944 : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e75a0f49cd546fab05842aca1c36fce : SUCCESS in 6m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a319da5cee184558b96bd187bcd410ac : SUCCESS in 29m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0b449cc617bc4eb9ab495f03f83aa0c7 : SUCCESS in 20m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/63be89af9c554209bf01c389bc176f2e : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":7},{"id":"4b0bb28f808d0b5d6b12478917868e73d416b386","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-27 23:52:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"95ba6d8912d8a339e22bc75feae6c20c13d5e950","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 23:57:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/205295f236ce4239b1b8315b3a352a3e : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a6d5d3537419483db0c642847fb9db45 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":7},{"id":"3b445e0152b142b04cbb8e26456aec7888b2047d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:14:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2fecb6ddc44fc390ec3bdb1e42e8c9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4b78103e4e643d88c0c066d5e642687 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4572d1b360684354ba816c2dfa573991 : RETRY_LIMIT in 4m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/274a39ecc65e4fd08d137712e97a959c : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"809bc8c9fb20f9d80d4d67241a25cf108069f31b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:29:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a9421c7f114cb0b0f57bae724a2944\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06599b2df8f74de580127eee74f9abb7 : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b1398d2bcf94b57b4017bbcfebbe8d3 : SUCCESS in 9m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34449d9ea10b44aa9a19211115c800b5 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/112952ad768f4cb5a9e88f951acda7dc : SUCCESS in 14m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ae48f85bcb40478738c5eb3323efd6 : SUCCESS in 14m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6385535978a345d0ba736b577318f8be : SUCCESS in 36m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/de50f71edfcd423cb8d23ca8686a8d6f : SUCCESS in 21m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a522d22c254543c2a8730cad89fcb6bc : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":7},{"id":"82437e7350657c11698dbdb5d3cdd310ca967b06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:29:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6cf783a9e4304e0a193aecab61417af8171ab557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 02:16:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f0e7b1753c4a729118b00156cb4e83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40fb09d0b3fb4edcabfe8a9b8e7a7d72 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5301eacf232742e6b113e22ffa6180cd : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4c0de1d8cd0498d9cbb2f5ca7ce253b : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355d5ca9db3c416392ed6f43bdffb21d : SUCCESS in 7m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92c0157e2cee405da0c3cfc463a72f8a : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":7},{"id":"3cd28db9118af232c417f5c3245fa8123bf745c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 02:16:23.000000000","message":"Change has been successfully merged\n\n6 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7},{"id":"7053c8f9a9b7e77bce6e56f11a4a7df1bfbd4fc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 02:21:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c63f04a25f6740eeb60cc095acff6b02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e943ed3862db432db1039b7cd6d8aab5 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":7},{"id":"220a9ed3a60b994a414b08fa5fd68acc7b9c382c","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-02-28 02:51:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-7-fc : SUCCESS in 49m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-03-966503-7-iscsi : SUCCESS in 1h 35m 5s","accounts_in_message":[],"_revision_number":7},{"id":"02a4a84c1f718d1579ac2cb505931147a5b47731","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-01 07:51:27.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_903/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":7},{"id":"bd00017bffd476899d3e859a08c167f5c1fd496a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-01 07:51:31.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/903/ : SUCCESS","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"741becdcda17c49d96c7e957103ab5e74516f2e0","revisions":{"7e9582953c88313c2575ee54650fd4f76b43223e":{"kind":"REWORK","_number":1,"created":"2025-11-09 00:19:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/03/966503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/1"}}},"commit":{"parents":[{"commit":"bc5591a1f9ceba64eaf5e5200b0bed60f79914ab","subject":"Hitachi: Add support for Hitachi VSP One B20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc5591a1f9ceba64eaf5e5200b0bed60f79914ab"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:18:30.000000000","tz":330},"subject":"WIP: Hitachi: Use override_config in tests","message":"WIP: Hitachi: Use override_config in tests\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e9582953c88313c2575ee54650fd4f76b43223e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e9582953c88313c2575ee54650fd4f76b43223e"}]},"branch":"refs/heads/master"},"161b1be751853ea139eac0946b6aad934ccb268a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-14 06:11:09.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/03/966503/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/2"}}},"commit":{"parents":[{"commit":"20abb0a43f96f534aceec25cb3fa27378019ae5d","subject":"Add Adaptive QoS setting based on volume size.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20abb0a43f96f534aceec25cb3fa27378019ae5d"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-14 06:11:09.000000000","tz":0},"subject":"WIP: Hitachi: Use override_config in tests","message":"WIP: Hitachi: Use override_config in tests\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/161b1be751853ea139eac0946b6aad934ccb268a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/161b1be751853ea139eac0946b6aad934ccb268a"}]},"branch":"refs/heads/master","description":"Rebase"},"61fc0bc3f7dfec7c291a49c1a48ceb3a6d14b68f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-14 10:04:49.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/03/966503/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/3"}}},"commit":{"parents":[{"commit":"64b10bef5a7cced572f872ea2e308c04a2e601a1","subject":"Add Adaptive QoS setting based on volume size.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/64b10bef5a7cced572f872ea2e308c04a2e601a1"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-14 10:04:49.000000000","tz":0},"subject":"WIP: Hitachi: Use override_config in tests","message":"WIP: Hitachi: Use override_config in tests\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61fc0bc3f7dfec7c291a49c1a48ceb3a6d14b68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/61fc0bc3f7dfec7c291a49c1a48ceb3a6d14b68f"}]},"branch":"refs/heads/master","description":"Rebase"},"4347efef7a441a3662c3a2442741242bab0dd243":{"kind":"NO_CHANGE","_number":4,"created":"2025-11-15 13:59:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"real_uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/03/966503/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/4"}}},"commit":{"parents":[{"commit":"2d89cc13035cdf01bf77e7bfb3827a55ed3a54ee","subject":"Hitachi: Add Adaptive QoS setting based on volume size.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d89cc13035cdf01bf77e7bfb3827a55ed3a54ee"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-15 13:59:48.000000000","tz":0},"subject":"WIP: Hitachi: Use override_config in tests","message":"WIP: Hitachi: Use override_config in tests\n\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4347efef7a441a3662c3a2442741242bab0dd243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4347efef7a441a3662c3a2442741242bab0dd243"}]},"branch":"refs/heads/master","description":"Rebase"},"ba156dd895c4b3458f2b8ea467e1668307ad52f5":{"kind":"REWORK","_number":5,"created":"2025-11-18 11:11:14.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/03/966503/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/5"}}},"commit":{"parents":[{"commit":"2d89cc13035cdf01bf77e7bfb3827a55ed3a54ee","subject":"Hitachi: Add Adaptive QoS setting based on volume size.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d89cc13035cdf01bf77e7bfb3827a55ed3a54ee"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2025-11-18 11:09:48.000000000","tz":0},"subject":"Hitachi: Use override_config in tests","message":"Hitachi: Use override_config in tests\n\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba156dd895c4b3458f2b8ea467e1668307ad52f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ba156dd895c4b3458f2b8ea467e1668307ad52f5"}]},"branch":"refs/heads/master"},"798b4fc29a02737eb703d4db3e52e0a471a1726d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-01-27 23:38:10.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/03/966503/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/6"}}},"commit":{"parents":[{"commit":"0bfe02ab5679ae6d87a06a1b6be7e8741acf7748","subject":"Hitachi: Add Adaptive QoS setting based on volume size.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bfe02ab5679ae6d87a06a1b6be7e8741acf7748"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-01-27 23:38:43.000000000","tz":-480},"subject":"Hitachi: Use override_config in tests","message":"Hitachi: Use override_config in tests\n\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/798b4fc29a02737eb703d4db3e52e0a471a1726d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/798b4fc29a02737eb703d4db3e52e0a471a1726d"}]},"branch":"refs/heads/master"},"741becdcda17c49d96c7e957103ab5e74516f2e0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-27 17:53:56.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"real_uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/03/966503/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/966503/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/966503/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/966503/7"}}},"commit":{"parents":[{"commit":"c757d57c229d0032e0c4943b0e6ee0963d1cc525","subject":"Merge \"Hitachi: Add Adaptive QoS setting based on volume size.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c757d57c229d0032e0c4943b0e6ee0963d1cc525"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-11-09 00:17:31.000000000","tz":330},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-02-27 17:53:56.000000000","tz":0},"subject":"Hitachi: Use override_config in tests","message":"Hitachi: Use override_config in tests\n\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\nChange-Id: I1a04ab75cdeb7935a5799960344efaec248c4979\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/741becdcda17c49d96c7e957103ab5e74516f2e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/741becdcda17c49d96c7e957103ab5e74516f2e0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
