)]}'
{"id":"openstack%2Fcinder~966544","triplet_id":"openstack%2Fcinder~master~Ifddf30821e8a6183f31059ea68278f6165583f16","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"37488":{"account":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"last_update":"2025-11-19 07:18:03.000000000","reason":"\u003cGERRIT_ACCOUNT_37488\u003e replied on the change","reason_account":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"}}},"hashtags":[],"change_id":"Ifddf30821e8a6183f31059ea68278f6165583f16","subject":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in _copy_from_remote_cache because `start_file_copy` request expects just only FlexVol name, but receives share_path with leading slash.","status":"NEW","created":"2025-11-10 13:03:20.000000000","updated":"2025-11-19 10:24:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cf7a55908c4e17e7e352fbced09c93cb66902422","_number":966544,"virtual_id_number":966544,"owner":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-19 10:24:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-10 13:04:12.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-10 13:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-10 13:58:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-10 16:00:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-10 16:01:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-10 16:48:07.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-10 18:05:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-12 18:11:27.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"434cf12d5ead9dc769100df0b37a2f0ad57ac117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"date":"2025-11-10 13:03:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6260a61b78649a064b6c2e116788c8158ecc3206","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-10 13:04:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"675e2adb2e61675de38218f80042341ab4d9cde2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 13:28:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f228c1f44d2c4180aa6025942d3b81e2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/220ed31ebcab407f97a28c191cda3be0 : FAILURE in 11m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8dc0485b0dcc4b80bfad796aa70495a0 : FAILURE in 15m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f07a325f43994c32ad670cc38f4aa1d7 : FAILURE in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30ad0d96888ac72aa04a60723004d646b4cba677","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-10 13:58:45.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifddf30821e8a6183f31059ea68278f6165583f16 .","accounts_in_message":[],"_revision_number":1},{"id":"60d1eacd32b97396968dcf5759de6d7ac2885dcd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-10 16:00:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d12c1aedaa924e3280237f192fc1bda1 : SUCCESS in 54m 19s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7fd7a0f336c2463d93106b176813304e : FAILURE in 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9c17198124129139ce205fc1282e62d5e797aed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 16:01:07.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75f47505c0374aaba499b80de8a2ace5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf836bbd6e64322a5c8b4595e9ba1a2 : FAILURE in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/279f940414024e64845a07e736db4264 : FAILURE in 11m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4815cf244e0e47c790828bb3c160985b : FAILURE in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6baefeacc8433a886036b8c12476c9 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/75508b0a1fe446deb64bbeb79f3c9671 : SUCCESS in 59m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc7a96fa8d70496b8af9c44356b91d4d : SUCCESS in 58m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/def7779f22914b0d96bc599ed3829122 : SUCCESS in 1h 18m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9772814c625542e284eb7cf7eb8a3d20 : SUCCESS in 58m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b07fde053086473691a30429b5baf2b5 : FAILURE in 21m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b709dd475f1e4599b8f57dcfec462efd : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50a04ecb7d564e02918e1f03f501ddd4 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abc2a676aeb440798e154b9156ad758e : SUCCESS in 5m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27bf18c6ef5941a695bca92e2e6edde9 : SUCCESS in 9m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99d6bec0362d49e59c52de21013920a2 : FAILURE in 39m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3079b46e55c41c983ea949a03005cf1 : SUCCESS in 4m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c6afc2fc903249b7941b24d6b357028a : SUCCESS in 1h 54m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/36ad451c086144a6afa1cb57108d311a : FAILURE in 2h 52m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76e0d441be9449c292fe5c6a4393be9d : FAILURE in 2h 34m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a5639351970a433bbe1afd830be4ecb1 : SUCCESS in 1h 52m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f36dc605daff48228149247eea4aba87 : SUCCESS in 33m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/468e0b26ad4a4f1a89995df36a9f1bb7 : SUCCESS in 53m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ed14431e171844b9bd9ed8336ca7f5b1 : FAILURE in 2h 08m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f3e86f55e574f9cab6c445aa2f2a74b : SUCCESS in 41m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d052cda62d6a4b21bfd8d02055927142 : SUCCESS in 1h 40m 36s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e8135f065d774b9c87284b5274eca130 : FAILURE in 1h 55m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b4233aaeed94234919fe86f6e4e8a6a : SUCCESS in 1h 52m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29f71b0e4f924d8b9071252d4a3aa4cb : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9f63cb027b01911cf380f7534a5afa91f6462266","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-10 16:48:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966544 : FAILURE in 1h 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3fd8b797be267168e15aa6116cbc69a2ba9e749f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 18:05:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2357/.","accounts_in_message":[],"_revision_number":1},{"id":"9bc8816055f1942bc91e4eaea78aa8521747c520","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-10 18:27:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2471/.","accounts_in_message":[],"_revision_number":1},{"id":"67efe071b4279d3a2d9c3450c7386c838a06f6cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"date":"2025-11-11 04:08:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d4b058b6668d36d301352c2295214cf074b0d8c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-11 04:09:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9e05ebd69a613af95c873ad1388c4e400e38b143","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 04:35:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67a2d153661406db9599105d85ddb79\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/943e22e7122d41b3ba5da1fb74d6b4a0 : FAILURE in 18m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7cf4de9b61f4c69ada58057f958ad7d : FAILURE in 18m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2bdfe607f9e742299e4e19a62daca30e : FAILURE in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe52e37ec72856f16d7bb4055b5d138e8f00c6d1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-11 05:05:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ifddf30821e8a6183f31059ea68278f6165583f16 .","accounts_in_message":[],"_revision_number":2},{"id":"c53a9396e5b6aa8c46af16d1d26ce19e68f5b220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 07:10:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/214a02a183904686ad02bccbe5171213\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9cfb7312c85402da0b64e9e9f704ef6 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ac9142a469c4d56814e65d8c40d9589 : FAILURE in 9m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71e21276fadc425f8062cb06fed81888 : FAILURE in 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33afafa90bad4f10b348b8d21b88f3d1 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/6d131f011e054aa88b75ad6cb6a09488 : SUCCESS in 41m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0255b25474f84358b618308a99ef0c1c : SUCCESS in 58m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/923b626e3d344ea69e628f97bc5ba5d1 : FAILURE in 56m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49ae08dec9e2461d9fb57a28ae07bc36 : SUCCESS in 52m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bbe596416ab9463cbfaa65f1ba0112cd : FAILURE in 25m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d8c21c6636242c29095e8a827fd5898 : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e074eaebaa554d4a8eaa33e20a256289 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4fb0fa204626412d89a1a6a094691c7d : SUCCESS in 5m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c5684cd330e4197a519f265b7403d6e : SUCCESS in 8m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91611c48dab6460682271f52574439b3 : FAILURE in 39m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a10a29f9f68f4b51b1cdedb517d9eeeb : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5fb63a9ceb06404b88c1a8370b35b460 : SUCCESS in 1h 52m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b8b74f997034793bceb3ca351d682d3 : FAILURE in 2h 57m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/481db33d69b14c028c1e6feeffa1964a : SUCCESS in 2h 06m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e94b6526ad8348bead81469a48946899 : SUCCESS in 1h 59m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/74eef210cfb847919705470505790914 : SUCCESS in 30m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1dd01fe42e2645578c429db4581932f6 : SUCCESS in 57m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a6409cde77504f468d3469668042b3cc : SUCCESS in 2h 09m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51e2abb1d9aa4ae093b7aa5a3f23ae40 : SUCCESS in 33m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/84d403c8971340a6a78fb42b122b4140 : SUCCESS in 1h 39m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8b1a36577bd8407a9004c4815b505ab9 : SUCCESS in 1h 41m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d78391bfa8804903a848339b40d39c12 : SUCCESS in 1h 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cfa1a73c3e14c6b804e6d71e3361443 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":2},{"id":"35c1279176f2398872b0e985be97609ac8f33b75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-11 11:07:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/202accb7950d46f5b44af1f2d31771d8 : SUCCESS in 46m 42s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5c84dfb0f37c4718b3127733815327a1 : FAILURE in 59m 59s","accounts_in_message":[],"_revision_number":2},{"id":"402dda5e2475f0a90f745bd54209e556552a98c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 11:46:37.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2491/.","accounts_in_message":[],"_revision_number":2},{"id":"3295f3afbbac3e4a4d3e49210c2bd3c9aefbb4d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-11 12:13:41.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2377/.","accounts_in_message":[],"_revision_number":2},{"id":"d57457776e139e5e95a55467d870ac11b8d05486","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-11 12:59:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966544 : FAILURE in 1h 11m 51s","accounts_in_message":[],"_revision_number":2},{"id":"36dd748fae28a7d13a1651fda34744a91ab31878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"date":"2025-11-12 03:43:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"40de279e60d17e34b0be76c3dd6b300c796dc462","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 03:44:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e32d986c762ce8eb938a91f4200a4a7ae9a42132","author":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"date":"2025-11-12 03:45:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ec537fddb1434ec1b00fedec3c89a66ca5cb94d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 04:07:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a84181b65c44629bca1c4720765bf9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99175fb0a4fe43abaf7b8c6c65427742 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1610b6a7da8e43ad9498e87c48dac4f7 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/077de28699e54bfca4d078e4ae750e2b : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51e86d7f4ca076226a418464e94a40ec3cfdfff0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 04:15:27.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2521/.","accounts_in_message":[],"_revision_number":3},{"id":"2e6dc8210d74ded0d597a797ad6a571a1998c2cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-12 04:15:27.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2407/.","accounts_in_message":[],"_revision_number":3},{"id":"4c78f06cb4a305e7eb2e69434349f87e85d2188b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 04:41:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifddf30821e8a6183f31059ea68278f6165583f16 .","accounts_in_message":[],"_revision_number":3},{"id":"bbeff2500ac17fe98ef568bb750801220012649a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 06:56:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/556805e465e441c69f35e2bdca018f4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f812ed6502144c4a6a7bf732cec8282 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c8ff655a4f4f2bbf62b6517460821a : SUCCESS in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/508af2bfe3284ef69e1a8b21163b5c5d : FAILURE in 23m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848802dff6f04c4f8a4de5d70353efe2 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/51094fa1379b4c4fbeeebfe29bc152b6 : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/079d54b387e4478a992aef72c1ac6c6a : SUCCESS in 48m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c24dae0a5224572ab31e3a6ba49b10e : SUCCESS in 1h 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21bc1ed743fc4e179671c90557a98f3f : SUCCESS in 48m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e8d0df81689d4987b4c9940227b00293 : SUCCESS in 20m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/80211b037b434fc8ac0c0413adcc168e : SUCCESS in 3m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b500caab61eb46458f7a530db4e50825 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66942706575444ccb0e0f05d9a55271b : SUCCESS in 6m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ad86d842d4624fc7919d9e5886b00119 : SUCCESS in 10m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/454607cde37447f39dca5c6fd25fbea3 : FAILURE in 21m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5ed921053c44cc4a81b6aab3be83e32 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e30d306b0207468aa2c9f2d37fb5bc19 : SUCCESS in 1h 53m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7a274802ef344865afb90250568bf4fd : FAILURE in 1h 46m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bcb555f5fd142ebbf555c7b81a0c6d0 : SUCCESS in 2h 19m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d8dfacd1567d489cb388b16bfacaaea4 : SUCCESS in 2h 00m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2b5e3c7e8d9c4febbe1daef72d84acbb : SUCCESS in 26m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/361beb4c52dd41d5956852bc18e91ce4 : SUCCESS in 58m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e1547e0c5804753b64c6869faef1f15 : SUCCESS in 1h 39m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f36d43d70b1746648b0b8cb06fbbdf09 : SUCCESS in 41m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/805c601ef0284691a4f86c8efb3aff35 : SUCCESS in 1h 31m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/006b2cac58444bf48bef94d3196996ba : TIMED_OUT in 3h 08m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1eb9aca7529948fab4430e9759c37afd : SUCCESS in 2h 05m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dfffcdc353945aaa39f275a5f2b8327 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":3},{"id":"8b09ed3f2bf12cb4d543b5b648bb5938b9e26e66","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-12 09:11:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966544 : FAILURE in 1h 12m 17s","accounts_in_message":[],"_revision_number":3},{"id":"79d0a4b8a7c23447492e9908bddafa566ecc8dfd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-12 14:36:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/15e49525c18f4a33b317dd8046b0adba : SUCCESS in 51m 07s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5e53a00fcd6242f393b7ef4187751354 : FAILURE in 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5a026ec0913b50f93e2c0d0edb512595ab916962","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-12 18:11:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-44-966544-3-fc : SUCCESS in 59m 14s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-44-966544-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4d6476983cd6f7a6edc25434f0c9facbdf50920c","author":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"date":"2025-11-19 07:18:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"495500534efea7d60fb2a4ce5d1af875c2dad6f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:48:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c3772b9d474601837ed3a4b675dff4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ee5c3cb340a4941ae5e27c207bd1e15 : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/076eab243f354ac79d59db325478d8e1 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e20c4fc19b934e0cb41daac8fbb16540 : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79a5393adb5e222aaaf05caca3e80b7669963488","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-19 08:29:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/79fc572012884509b7a28dfd9254d2fe : SUCCESS in 56m 54s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/675f4d180a614227923675e7f09336b7 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cf7a55908c4e17e7e352fbced09c93cb66902422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:24:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf14dc06a6cb4ba19132bf057bde067e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fea11fbaa94f4587726a88caa16936 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/706ea5ac15894cc2ab159ec5bb262c0b : SUCCESS in 12m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/848623695bd44422923d7891d0146276 : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a200678330744ad8b57aedaf35661f6a : SUCCESS in 14m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b750a86eac864f2bb5aff8c7d5589787 : SUCCESS in 49m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2bd81d05063042fcaf11ce48dff0d54c : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d90b789f5d8c45099770ed5455d012ba : SUCCESS in 1h 21m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c73eb07e266d4aa9b1a662445eccb680 : SUCCESS in 1h 01m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f0a701bb74184e1f95b3c884946ecb46 : SUCCESS in 21m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13cb0bf0acc043d6a0f660a815ccdbb8 : SUCCESS in 4m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/011a03728619478987fc4c16a257c9e2 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/149e22bdecf141a0ad04a87117340ab6 : SUCCESS in 5m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/07d27de0c04846b7909ac46c523ffcc3 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/154e307e984b475ebe971d520c5e1b37 : FAILURE in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8a6fbbf93ac47adb5785a6df4a8e477 : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1de8ca9c0a08421fa1942b40fff70a67 : SUCCESS in 1h 45m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6fb7e65d02f841ccab852545a33d223a : FAILURE in 2h 34m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c4fd4198e0b4d9f9f57c4efb7749d4a : SUCCESS in 2h 28m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/045946ad213045f4a11781dd9e34fca8 : SUCCESS in 1h 52m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3afab07b0eb743bda1447e3d8cc729bf : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83bf54be7837435d8e49a5d3d52b2f8c : SUCCESS in 34m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b37b086e46e4c23a5924c87bffd98d2 : SUCCESS in 1h 35m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5aca9c099582403299587d40f6c83a2c : SUCCESS in 47m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/752540862eb94f498e95e6b60fe6eff5 : SUCCESS in 1h 45m 35s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/06d14b60d20b43e686e9e5d73af2d23a : FAILURE in 2h 55m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6383b8a45c2452883e5c9db8690e621 : SUCCESS in 1h 03m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e58c0c6d3dc84956922145b5a85340fb : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed565b9a47df5336b9f99f9e4cf0d333b0aa8896","revisions":{"d14d9b9f533fe8d982ee2e3d6aff928777ce99e8":{"kind":"REWORK","_number":1,"created":"2025-11-10 13:03:20.000000000","uploader":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"ref":"refs/changes/44/966544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/966544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/966544/1"}}},"commit":{"parents":[{"commit":"f76fed0d8e5b7baf701583557d7104dc2422432b","subject":"Merge \"Fix: LVM optimized migration path during retype\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f76fed0d8e5b7baf701583557d7104dc2422432b"}]}],"author":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-10 13:01:44.000000000","tz":540},"committer":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-10 13:01:44.000000000","tz":540},"subject":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in _copy_from_remote_cache because `start_file_copy` request expects just only FlexVol name, but receives share_path with leading slash.","message":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in\n_copy_from_remote_cache because `start_file_copy` request expects just\nonly FlexVol name, but receives share_path with leading slash.\n\nThis change removes leading slash from share_path before passing to\n_copy_file method.\n\nClose-Bug: #2131010\nChange-Id: Ifddf30821e8a6183f31059ea68278f6165583f16\nSigned-off-by: TOMOFUMI-KONDO \u003cugax2kontomo0314@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d14d9b9f533fe8d982ee2e3d6aff928777ce99e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d14d9b9f533fe8d982ee2e3d6aff928777ce99e8"}]},"branch":"refs/heads/master"},"ad25b0d1eecf191aa55abd4e1ad146465e275278":{"kind":"REWORK","_number":2,"created":"2025-11-11 04:08:10.000000000","uploader":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"ref":"refs/changes/44/966544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/966544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/966544/2"}}},"commit":{"parents":[{"commit":"f76fed0d8e5b7baf701583557d7104dc2422432b","subject":"Merge \"Fix: LVM optimized migration path during retype\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f76fed0d8e5b7baf701583557d7104dc2422432b"}]}],"author":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-10 13:01:44.000000000","tz":540},"committer":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-11 04:07:56.000000000","tz":540},"subject":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in _copy_from_remote_cache because `start_file_copy` request expects just only FlexVol name, but receives share_path with leading slash.","message":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in\n_copy_from_remote_cache because `start_file_copy` request expects just\nonly FlexVol name, but receives share_path with leading slash.\n\nThis change removes leading slash from share_path before passing to\n_copy_file method.\n\nClose-Bug: #2131010\nChange-Id: Ifddf30821e8a6183f31059ea68278f6165583f16\nSigned-off-by: TOMOFUMI-KONDO \u003cugax2kontomo0314@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad25b0d1eecf191aa55abd4e1ad146465e275278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ad25b0d1eecf191aa55abd4e1ad146465e275278"}]},"branch":"refs/heads/master"},"ed565b9a47df5336b9f99f9e4cf0d333b0aa8896":{"kind":"REWORK","_number":3,"created":"2025-11-12 03:43:32.000000000","uploader":{"_account_id":37488,"name":"Tomofumi Kondo","email":"ugax2kontomo0314@gmail.com","username":"TOMOFUMI-KONDO"},"ref":"refs/changes/44/966544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/44/966544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/44/966544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/44/966544/3"}}},"commit":{"parents":[{"commit":"f76fed0d8e5b7baf701583557d7104dc2422432b","subject":"Merge \"Fix: LVM optimized migration path during retype\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f76fed0d8e5b7baf701583557d7104dc2422432b"}]}],"author":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-10 13:01:44.000000000","tz":540},"committer":{"name":"TOMOFUMI-KONDO","email":"ugax2kontomo0314@gmail.com","date":"2025-11-12 03:43:18.000000000","tz":540},"subject":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in _copy_from_remote_cache because `start_file_copy` request expects just only FlexVol name, but receives share_path with leading slash.","message":"cinder-volume ONTAP: Remove prefixing slash of FlexVol share_path in\n_copy_from_remote_cache because `start_file_copy` request expects just\nonly FlexVol name, but receives share_path with leading slash.\n\nThis change removes leading slash from share_path before passing to\n_copy_file method.\n\nClose-Bug: #2131010\nChange-Id: Ifddf30821e8a6183f31059ea68278f6165583f16\nSigned-off-by: TOMOFUMI-KONDO \u003cugax2kontomo0314@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed565b9a47df5336b9f99f9e4cf0d333b0aa8896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ed565b9a47df5336b9f99f9e4cf0d333b0aa8896"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
