)]}'
{"id":"openstack%2Fcinder~966924","triplet_id":"openstack%2Fcinder~master~I53dba1f81a46cd75486aae00b70a1567bf1f6927","project":"openstack/cinder","branch":"master","topic":"v-c-ionice","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-11-13 15:17:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53dba1f81a46cd75486aae00b70a1567bf1f6927","subject":"docs: update data leakage docs","status":"MERGED","created":"2025-11-12 19:36:16.000000000","updated":"2025-11-13 15:20:42.000000000","submitted":"2025-11-13 15:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966924-v-c-ionice","meta_rev_id":"dd4f4cedf3dab3f8251334c097831e2e915e9a70","_number":966924,"virtual_id_number":966924,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-13 15:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-13 14:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-13 14:36:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 19:37:01.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-12 19:57:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-12 20:34:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-13 01:28:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-13 05:13:15.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-11-13 05:58:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-13 14:36:28.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"174f46695a2b4a1a911f0eade02bc697998788cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-12 19:36:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30482695e5af3c8445d74bfaf252c373545d3a4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-12 19:37:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a7bbbdc7cf6aae7c4cc9d6cfdfd5918cb14475aa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-11-12 19:39:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a1e286c905ccb0a02566954a05bdaa7751c39c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:57:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea21f488655433593182c807190061a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf9202c247d4894a8f305930172bf37 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a321110ca454566a5fe78a002de2ad0 : SUCCESS in 15m 08s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/afab2aec97ab419b811769d64f548373 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4516aca93769babb9a0e8eb33cd0773721bdc02d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-12 20:34:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I53dba1f81a46cd75486aae00b70a1567bf1f6927 .","accounts_in_message":[],"_revision_number":1},{"id":"93fdb7a7f8c8e87f24f97033a880c9b001951fd9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 01:28:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2544/.","accounts_in_message":[],"_revision_number":1},{"id":"03925869c656c0a3d8242d93a5aa76195866e906","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 01:29:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2430/.","accounts_in_message":[],"_revision_number":1},{"id":"fe4f2d0045a9ebef6d7e0cc551ddc15e019d6964","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-13 05:13:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-24-966924-1-fc : SUCCESS in 1h 5m 41s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-24-966924-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"117eb7693468fd7b4c65c2ba5dd3e38233b0880e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 05:58:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966924 : FAILURE in 1h 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a0507bbd4453e285bda948eca805c15740cc321d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-13 14:36:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"566c06dce9fba37c59c354be7d4cc99a510a0686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 14:36:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87f9c43ec7bc895aa301cb3cf45c6d6acf7f3854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:17:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68efdc1bc5ca4c4ba7ed88abf7f31414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c6c584b3e8423792a3fdfd4be44cdf : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cd9445f9454581acea28e71ab26797 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b1836a7c04b2770437347e0942bb120880927c63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd4f4cedf3dab3f8251334c097831e2e915e9a70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 15:20:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1873b60ee1904b7e89da7822b467974a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07549444176b4be9a21ba7171c95bc04 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ca07c474adffc97b6f6679aafc1abecd0df4dd2","revisions":{"9ca07c474adffc97b6f6679aafc1abecd0df4dd2":{"kind":"REWORK","_number":1,"created":"2025-11-12 19:36:16.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/24/966924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/966924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/966924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/966924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/966924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/966924/1"}}},"commit":{"parents":[{"commit":"017d2283cfcf2d68aac089dfdc3e25a28fd847d9","subject":"Merge \"api: Remove AdminController controller\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/017d2283cfcf2d68aac089dfdc3e25a28fd847d9"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-11-12 19:34:39.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2025-11-12 19:34:39.000000000","tz":-300},"subject":"docs: update data leakage docs","message":"docs: update data leakage docs\n\nThe volume_clear_ionice option isn\u0027t currently mentioned in the\ndata leakage documentation, so add it.\n\nChange-Id: I53dba1f81a46cd75486aae00b70a1567bf1f6927\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ca07c474adffc97b6f6679aafc1abecd0df4dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ca07c474adffc97b6f6679aafc1abecd0df4dd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
