)]}'
{"id":"openstack%2Fcinder~966986","triplet_id":"openstack%2Fcinder~master~I02c1cf40a1f7390a06f8eb2300670e0bc06bfc0f","project":"openstack/cinder","branch":"master","attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-11-13 06:35:54.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I02c1cf40a1f7390a06f8eb2300670e0bc06bfc0f","subject":"[DNM] Add `is_glance_image_volume` field to Volume object","status":"NEW","created":"2025-11-13 04:49:55.000000000","updated":"2025-11-13 08:56:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"a9b30d26def105daf59425b0f0a84cfe219871ff","_number":966986,"virtual_id_number":966986,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-13 06:27:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-11-13 08:56:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-13 04:50:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-13 05:20:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-13 05:25:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-13 05:27:39.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-13 06:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-13 06:35:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-13 06:48:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-13 08:56:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"1f8d2824eac67a01147a6c2c84131b9ebd26f95b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2025-11-13 04:49:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51d853cb74cf375377868e7ae8396937e20a8dd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-13 04:50:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"f98e56df65d542ca7869ffb5058fad4b5212d10e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 05:20:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2304ee5a875e4eeaa5e4ac0bcdca585b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ab2303ca7a14b06a4934cfd5242f873 : FAILURE in 14m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f9ef1622550d4dc080e3077d65cca244 : FAILURE in 23m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb10e980657b475aab9aedd019c151f9 : FAILURE in 28m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12ed9b3468c71d3d9c3705c0cf2d3fa199044589","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 05:25:24.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2545/.","accounts_in_message":[],"_revision_number":1},{"id":"2d8553da548522c29a94146544ad471f11e5395a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-13 05:25:29.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2431/.","accounts_in_message":[],"_revision_number":1},{"id":"d221d0306111b73a4c6576c32e260bc377943098","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-13 05:27:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I02c1cf40a1f7390a06f8eb2300670e0bc06bfc0f .","accounts_in_message":[],"_revision_number":1},{"id":"ca794e69e8866d7bbc895a751c272711a404f404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 06:27:40.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddad97f60b294fa0856a32a0ce00581a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d03e75a5c54a82a123cbaf97f7e06e : FAILURE in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650f794475b145458d86532f12d336da : FAILURE in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4165baaa123f4ea6a51b696fa4c08e0a : FAILURE in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9488b428cb6849e89db6bf839e76c932 : SUCCESS in 13m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c9ac01f352ea43b78c0f17f0364b9dda : FAILURE in 36m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f0ed35de4c2446bbf4da35b146851dd : FAILURE in 27m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d337df4880349eda0b5fcd3988c1bbf : FAILURE in 1h 21m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/098676c452e043dda402b35e07de479b : FAILURE in 55m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7131e07bb712405d853aa9d8c79d1ac3 : FAILURE in 23m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f9f3032005484f689a213edda9ea24e3 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/208b96f17ce041268ae2e418b24efd98 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b41c1fdfbd24bfeaca23aa637b3fd42 : FAILURE in 4m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/60841a0b1aae4440bd5dc468e4bb1da5 : FAILURE in 9m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/90eafce27fe54f4cbb45f3fb6218a3a2 : FAILURE in 34m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/542bbc59e18e45c4a9350e7283a90f22 : SUCCESS in 4m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5af5e20c378643a6a33c4a60b108b391 : FAILURE in 4m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3b09e204cd5472ea15446611332132c : FAILURE in 1h 08m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3bd6b8a0e6414b01919a49c6bf292c52 : FAILURE in 4m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f4384144788c43d39d8ab6e7a8bb9538 : FAILURE in 1h 30m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b53cd51386384ea29119515865a51916 : FAILURE in 17m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0d8fff00521c4362a66d07805e66f0cf : FAILURE in 32m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0b0c4e8f1204f2dbb990143ecf77b50 : FAILURE in 1h 19m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/262f854cfaf2486d99cd5e9e54ceb94c : FAILURE in 26m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2990cfa218ce437eaf0c78c3684a1fd2 : FAILURE in 1h 26m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ec373e7b1f5a43d2bb23d5e18746cccf : FAILURE in 1h 24m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7dceba0e11bc4f54ae6504e77f230c12 : FAILURE in 1h 08m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/451eef6aa38d4b76811811103316e333 : FAILURE in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0988200ba9c86570cf1414cbe63943b9b6c85c99","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-13 06:35:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/966986 : FAILURE in 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"418ff6e8c22f815d613198e6ac93f624bd840abb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-13 06:48:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e2f66e9c0b8746b38194387613ae8ac3 : FAILURE in 14m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9174ec098dcb4b5e896ae817813a2341 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a9b30d26def105daf59425b0f0a84cfe219871ff","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-11-13 08:56:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2a806414562fa3d15d158830eed970c8530643e","revisions":{"b2a806414562fa3d15d158830eed970c8530643e":{"kind":"REWORK","_number":1,"created":"2025-11-13 04:49:55.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/86/966986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/86/966986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/86/966986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/86/966986/1"}}},"commit":{"parents":[{"commit":"4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe","subject":"NFS driver: Fix driver fails to verify virtual size for Glance with Cinder/NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a84a60be3d1c144e36d5c9ed3d4d82aff8a7fbe"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-13 04:40:16.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2025-11-13 04:40:16.000000000","tz":-180},"subject":"[DNM] Add `is_glance_image_volume` field to Volume object","message":"[DNM] Add `is_glance_image_volume` field to Volume object\n\nChange-Id: I02c1cf40a1f7390a06f8eb2300670e0bc06bfc0f\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\nAssisted-By: Cursor (claude-4.5-sonnet)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2a806414562fa3d15d158830eed970c8530643e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2a806414562fa3d15d158830eed970c8530643e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
