)]}'
{"id":"openstack%2Fcinder~967140","triplet_id":"openstack%2Fcinder~master~Ib00e4b72b708287f5ef9c4cab03cf097c5401f22","project":"openstack/cinder","branch":"master","hashtags":[],"change_id":"Ib00e4b72b708287f5ef9c4cab03cf097c5401f22","subject":"Add support for adaptive QoS based on volume size.","status":"ABANDONED","created":"2025-11-14 00:52:33.000000000","updated":"2025-11-14 03:06:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c866db117d0019168042fe623007d9925792a329","_number":967140,"virtual_id_number":967140,"owner":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 00:53:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 01:49:51.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 01:53:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"22b82b5d2cf2932dd9fae9f481e5db9f36e26431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 00:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"857c1343025d008a927439f3dff7b554dfd78b0a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 00:53:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"75755cfc4cff779e2083908ecefed0641ff8d895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 00:54:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9252a7a7fffb731c7677138b08ac342b1e436429","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 00:55:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a2ec39ab733e200f1855428395b4a1536e238767","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 00:59:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bfea42f84eaaa0dd172908366f4dfc4fb4a1e01","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-14 01:00:07.000000000","message":"Abandoned\n\nWindows Newlines (oops)","accounts_in_message":[],"_revision_number":2},{"id":"d55b43df73255da48a74236a0aae210cc8890e27","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 01:49:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib00e4b72b708287f5ef9c4cab03cf097c5401f22 .","accounts_in_message":[],"_revision_number":2},{"id":"4dbbb6fd6b0ccd02b8bbbe866a92aa9f577eb00e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 01:53:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2480/.","accounts_in_message":[],"_revision_number":1},{"id":"0b32a402e173e3c7409ddf44efe987a05f001e50","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 01:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2594/.","accounts_in_message":[],"_revision_number":1},{"id":"59d212e5aa593706b77db088c7d2982af7d7c6db","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 01:59:22.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib00e4b72b708287f5ef9c4cab03cf097c5401f22 .","accounts_in_message":[],"_revision_number":1},{"id":"ea8f9947b9c9b176478510e6212d36b8e55e0412","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 03:02:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2481/.","accounts_in_message":[],"_revision_number":2},{"id":"c866db117d0019168042fe623007d9925792a329","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 03:06:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2595/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08244205c0e8e4f344776b06f89bfa8c7e62589a","revisions":{"4212d8fd179fb26bca0edfb30380df6fa27c6916":{"kind":"REWORK","_number":1,"created":"2025-11-14 00:52:33.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/40/967140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/967140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/967140/1"}}},"commit":{"parents":[{"commit":"047e717b981f360ffb949b958041af009965ed4f","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/047e717b981f360ffb949b958041af009965ed4f"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2025-11-13 16:02:18.000000000","tz":-480},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2025-11-13 16:02:18.000000000","tz":-480},"subject":"Add support for adaptive QoS upperIops setting based on volume size.","message":"Add support for adaptive QoS upperIops setting based on volume size.\n\nChange-Id: Ib00e4b72b708287f5ef9c4cab03cf097c5401f22\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4212d8fd179fb26bca0edfb30380df6fa27c6916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4212d8fd179fb26bca0edfb30380df6fa27c6916"}]},"branch":"refs/heads/master"},"08244205c0e8e4f344776b06f89bfa8c7e62589a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-14 00:54:59.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/40/967140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/967140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/967140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/967140/2"}}},"commit":{"parents":[{"commit":"047e717b981f360ffb949b958041af009965ed4f","subject":"Hitachi: support extending volume has snapshot(s)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/047e717b981f360ffb949b958041af009965ed4f"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2025-11-13 16:02:18.000000000","tz":-480},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2025-11-14 00:54:59.000000000","tz":0},"subject":"Add support for adaptive QoS based on volume size.","message":"Add support for adaptive QoS based on volume size.\n\nChange-Id: Ib00e4b72b708287f5ef9c4cab03cf097c5401f22\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08244205c0e8e4f344776b06f89bfa8c7e62589a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08244205c0e8e4f344776b06f89bfa8c7e62589a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
