)]}'
{"id":"openstack%2Fcinder~967145","triplet_id":"openstack%2Fcinder~stable%2F2025.1~I3214d445221d5e1fe87e5491e83e82c9934e8e26","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"27488":{"account":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"last_update":"2025-11-17 02:01:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3214d445221d5e1fe87e5491e83e82c9934e8e26","subject":"Merge \"Hitachi: Add support for Hitachi VSP One B20\"","status":"ABANDONED","created":"2025-11-14 01:47:23.000000000","updated":"2025-11-17 02:01:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4ab7e3a577ad15a86d7ac4684b7ae1caff20210","_number":967145,"virtual_id_number":967145,"owner":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-14 03:56:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":1,"date":"2025-11-15 12:49:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 01:48:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 02:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 02:42:19.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 03:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-14 05:18:20.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-14 06:45:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-15 12:49:14.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"}],"messages":[{"id":"acdbc73bdccfea8f87aaf7dce8915dfd1ec002df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-14 01:47:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"833fa7786a731207efa88ccb3385ee30c02b6bcd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 01:48:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"859af2ee83d81be5b4a4e08c6d8a8bd8e40631ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 02:08:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c28a6d43a176410caf51b82dddee92ef\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0facc6d992cb410f91ff7aaf4dce8bc3 : FAILURE in 15m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/980487388d86482293cd87514e69908f : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"166845de49a0c49dcc2f82d30dd660516bbe66d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 02:42:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3214d445221d5e1fe87e5491e83e82c9934e8e26 .","accounts_in_message":[],"_revision_number":1},{"id":"4241af17ee7f49e57d380a8b87f031360ee77c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 03:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa85d6c9dd504c06b0c7de78dd516c3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd538376af014e028539036e77c79e07 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d28fd7f750c458d9d74366259c9f372 : SUCCESS in 13m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/027f86ec5eaf4e97a94d1b5ced63732d : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94c38f45a4044c2b58a135a4ef65364 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/4aae991896a042feba65c493839a4cf8 : SUCCESS in 45m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f342be51feaf468399af024316e20fb7 : SUCCESS in 52m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ec333900dc240dc9e1c8c4f6eb6a9c6 : SUCCESS in 1h 05m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7045e192c70e46a78ebc60441bd2fc5e : SUCCESS in 40m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c11123b348e49db97e7e847f06cecb2 : SUCCESS in 11m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2da6952a11a24dc1be9e32f756126710 : SUCCESS in 16m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a0f99204fdae42b5bde1b70c4cce8b7e : SUCCESS in 4m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/347eda7f2b2e457bb243738f89bb0b87 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1dd66bb25f5d4bbaae8eecefcf600c71 : SUCCESS in 7m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f412040bfe5a4e4d8c8153ab8b81bf0d : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5afffb533aed443dadb4c376147b2151 : FAILURE in 21m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5188cefb83704d13a0554e9422bac3b6 : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d91562229d04ab6953461fe4fb2f807 : SUCCESS in 1h 10m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b1d38a3e49742f7b5a80e2d370ae7f2 : SUCCESS in 1h 21m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68d43eaac887442cbf6534bb997e4ae1 : SUCCESS in 2h 07m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e2b8a981c5de4e5ea6591cb59c97c264 : FAILURE in 16m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2f0639bb9e7747f4b74b4e4e911dee07 : SUCCESS in 37m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/740d948b70e943ce8cd176d45feec4e0 : SUCCESS in 1h 04m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b9e1014f1770499e97e07013fe3b6cde : SUCCESS in 1h 16m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4c93cedde5884675a7e70f62e7ccaa64 : SUCCESS in 34m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f447798d3ec44863baf6b4f726776090 : SUCCESS in 1h 37m 40s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0c0697a4970e440393429c521a89c5aa : FAILURE in 36m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6649b229e3cd48bfbcf32d34a0d03cc1 : SUCCESS in 1h 40m 41s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e70ab3bda6f487c9899b5f3ee59d707 : SUCCESS in 1h 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7276d3af1f3b419caefcf77a135a07dc : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"96de88686d8441eeeaacde342dd1084976d26191","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 05:18:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2597/.","accounts_in_message":[],"_revision_number":1},{"id":"7f71f1d9a8f79ba418980efc579812d1353ac07e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 05:57:32.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2483/.","accounts_in_message":[],"_revision_number":1},{"id":"1c5940a7c0154e966fd3466c8fea21d1eb7e7c6d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-14 06:45:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/967145 : FAILURE in 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a59c34d976df27258ca0db78dcdf49dcf1b2b75f","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-15 12:49:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4ab7e3a577ad15a86d7ac4684b7ae1caff20210","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-17 02:01:45.000000000","message":"Abandoned\n\nuse another change","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cb6a9b1aef1770694c7eb80bb34a1753458837d","revisions":{"8cb6a9b1aef1770694c7eb80bb34a1753458837d":{"kind":"REWORK","_number":1,"created":"2025-11-14 01:47:23.000000000","uploader":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"ref":"refs/changes/45/967145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/45/967145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/967145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/967145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/45/967145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/45/967145/1"}}},"commit":{"parents":[{"commit":"0c0c5bf5ce1a2104eef20a5504b37d09b750e3a0","subject":"Merge \"[NetApp] Enabling total_volumes capability support\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c0c5bf5ce1a2104eef20a5504b37d09b750e3a0"}]}],"author":{"name":"Zuul","email":"zuul@review.opendev.org","date":"2025-11-11 17:24:33.000000000","tz":0},"committer":{"name":"sapd","email":"saphi070@gmail.com","date":"2025-11-14 01:47:09.000000000","tz":420},"subject":"Merge \"Hitachi: Add support for Hitachi VSP One B20\"","message":"Merge \"Hitachi: Add support for Hitachi VSP One B20\"\n\nChange-Id: I3214d445221d5e1fe87e5491e83e82c9934e8e26\nSigned-off-by: sapd \u003csaphi070@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb6a9b1aef1770694c7eb80bb34a1753458837d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8cb6a9b1aef1770694c7eb80bb34a1753458837d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
