)]}'
{"id":"openstack%2Fcinder~967146","triplet_id":"openstack%2Fcinder~stable%2F2025.2~Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14","project":"openstack/cinder","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"27488":{"account":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"last_update":"2026-01-02 18:48:15.000000000","reason":"Change was submitted"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2025-11-18 06:23:47.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-01-02 18:48:15.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-12-01 21:57:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-01-02 18:48:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14","subject":"Hitachi: Add support for Hitachi VSP One B20","status":"MERGED","created":"2025-11-14 01:49:35.000000000","updated":"2026-01-02 23:27:14.000000000","submitted":"2026-01-02 18:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967146","meta_rev_id":"66236283cf315e631ea3c39b802cdbd0a1b5241d","_number":967146,"virtual_id_number":967146,"owner":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-02 18:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-01-02 16:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2025-11-18 05:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2025-12-01 21:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-12-19 09:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2026-01-02 16:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 01:49:35.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"state":"CC"},{"updated":"2025-11-14 01:49:35.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2025-11-14 01:50:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 02:49:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 03:56:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-14 06:20:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-14 07:23:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-17 01:47:26.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-11-18 05:00:26.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-12-03 08:45:37.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-12-03 08:46:09.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-19 09:45:04.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-01-02 16:35:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-02 19:35:58.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"f67c2729e74eed5080b983b5199bbc8080d650da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-14 01:49:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b036d10877709def0d05328a3920e241717de71","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 01:50:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2c55bcaf11d456cacc556e1dee3e79257fe54040","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 02:49:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14 .","accounts_in_message":[],"_revision_number":1},{"id":"59c270c835c5b46bf90b6418b5eb45167fd7e35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 03:56:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb4ec6ea36045249be79b8934a4bab0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d026e76686524d4abbf0b07109be4895 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c51f866c20c4dbc850f11c1e6007367 : SUCCESS in 9m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dcb86b9969a4f8babb397d80d6b82c3 : SUCCESS in 12m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1ddaeecc0d3461295c8adc7122aaabc : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31afd5581b66470a9c17c227aeb1b072 : SUCCESS in 16m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/5e874790483148b1bb5ab47c2e87e8f4 : SUCCESS in 57m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9981b83ee4d54496834363d799eda895 : SUCCESS in 52m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d54565e9526a4c6896ff2aaae6592f05 : SUCCESS in 1h 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1df648075fde40f2949dc597ceef9906 : SUCCESS in 1h 06m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b13df96307de4e418c3185d443b4e2b5 : SUCCESS in 16m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/071f68d7c4724c4bb3db8f6383618118 : SUCCESS in 24m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/264fbd680e064a14912fd043a60233aa : SUCCESS in 5m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a218db9cb327466ebb5465709d85faf6 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3ec9dc3f65e487a82e75b539a65bfff : SUCCESS in 5m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/71edcca98bf8434daa857003ddaee512 : SUCCESS in 9m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0275be7403bc4ceabefc6a0ab1e6641a : FAILURE in 26m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfd20772f313426898bfe0ae4de48533 : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc84d477bbd342f2b26a1a6fa4299441 : SUCCESS in 2h 00m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b904d45e24f4ff996de95dfb88d1816 : FAILURE in 1h 47m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/633aa1eefdf543a1a02ce3cb89a6fb90 : SUCCESS in 1h 43m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e4951573062542a58e2ff5dddf355e99 : FAILURE in 8m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/34442296c65641858150ceb1b6812b69 : SUCCESS in 29m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8adaeee84e94411f8cf51cfc6f6f8507 : SUCCESS in 1h 09m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/53bcf8742f4346a0b6d9f81f72334e24 : SUCCESS in 1h 44m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0beb6077ceb34315aebc03bfd2898eb1 : SUCCESS in 30m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81ac1f90a7e941bbbb870fe576e4ffa0 : SUCCESS in 2h 02m 22s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/45b6367480144f75a3f2a9718743508f : FAILURE in 17m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e0049a025ce4a248a6f62d9ee9771c1 : SUCCESS in 1h 47m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de72187616d549149f9327388461aea3 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e78401f9d8daf6785dcf5911aed8b4a77d228f67","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 06:20:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2598/.","accounts_in_message":[],"_revision_number":1},{"id":"f9f02fd542d3c8c5fbf4dc701fd83067112f61c1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 06:21:26.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2484/.","accounts_in_message":[],"_revision_number":1},{"id":"5d143bbc73e20ca28346ef8679283d9d82f8285b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-14 07:23:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/967146 : FAILURE in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"32d6cb74b4ae9eee78a80d7206a796402a463f62","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-15 12:44:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6040e94187b29929e1d6da4374b2e3966297f4b5","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-16 02:28:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"010fe95fcc92ab04eeb30f0302a18a8464d1ee81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-16 02:44:49.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"615bae74bcf385367723443fdf8f348d2bc3e572","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-16 02:45:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b9f2a7b717fbd7b40555b8b7226a435021398913","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-16 03:21:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/967146 : FAILURE in 35m 07s","accounts_in_message":[],"_revision_number":2},{"id":"cc62e594972c946cfcc1d1366184d498c2da0feb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-16 03:44:13.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14 .","accounts_in_message":[],"_revision_number":2},{"id":"7644707624ea3ad10f71d649dcc51bba86d9bf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-16 05:49:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14e785cb1a9d4c08b0e28cd057cc1f0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00367de7fcb748f78b1395dc111be755 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9636b719a6b84d0597053dbbf6c0b5fd : SUCCESS in 10m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/505709c3224a46aab163c0b1b3223680 : SUCCESS in 13m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4485fc35dc5b4714928e9dde7c762f00 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da7869a91644293a90d3244f8bf75d4 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/a367e69ccfa547c4a55d61f3fba9b3fb : SUCCESS in 33m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f59ced9d451642db917b2a3df5d7b978 : SUCCESS in 59m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d5ff310d9a945d8aae159f0c2f756f4 : SUCCESS in 1h 33m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42a7d2b2c7164320870a0f677bfd347e : SUCCESS in 50m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037ee67d1fb94dbc8d32cdb679d73d06 : SUCCESS in 16m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4193ba1033184027985debc215257729 : FAILURE in 18m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c321d61ea02842318f5a61049c1bdb0e : SUCCESS in 3m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/687053ab985a4825a21b95c84810f9c7 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ffbd0de1c8349228326a1b740cfd671 : SUCCESS in 5m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/34ab0ea2d93e49fe93f1506ee69d375f : SUCCESS in 11m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f320a6d3dade4c6295a67bacc1e13c5a : FAILURE in 35m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b296eefb6a004fab93bb6a2a914aa9ed : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a62a18bc70d74c66ac0f28c84af49d7f : FAILURE in 1h 43m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/abfc27fd4c6e415681d63c149ddcb5f5 : FAILURE in 3h 01m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ffb7d271b3c84f51b0b410b406a5186f : SUCCESS in 1h 46m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/96d17ecba22a4cbe95ae0d0b3c250f55 : FAILURE in 14m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e21339ccfd6d49cb93be969efbd29e42 : SUCCESS in 34m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/743e346579614f63b19af30e5d3ac3af : SUCCESS in 40m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d6ee658c9428417b8ae8ebb3fd05a35e : SUCCESS in 1h 44m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/964fbaa1b1344b73925d8ae5156fb087 : SUCCESS in 44m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c29650154a614570937f6e26c416e2b6 : SUCCESS in 1h 40m 11s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b02c99b90ae5486f802dbcd167423f88 : FAILURE in 8m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae909f5909c44887a4301d07af09ccf4 : SUCCESS in 53m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28a55ff0e7454a7bbbf26c50e9de819e : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":2},{"id":"60a19921b01dac47ac9ceb3d62dbc41f4d56d006","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-17 01:47:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17ac7af37236c640b4ab0bb6ef1db556237996ea","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-18 01:59:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05893e038620f58363cefda94b56f365c6c21b3d","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-18 04:26:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87fff1585c14995e1053fc98fafe5d6ca9453ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 04:48:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b6d43cafec430099f05cf95bc019e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50efc4ad3f96406ea362ceabc329c8d8 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74bb5e7c6f474909a618a344f027817a : SUCCESS in 13m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e5745b848ee4a65a57168a77adc7722 : SUCCESS in 8m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ed6794259c141c396816f1ee64db484 : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81d6dc4058e48039ef381211ec23e87 : SUCCESS in 15m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/9fa932e7638842e7a3d9e87ab77fd861 : SUCCESS in 59m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3440975f138d478a94675e312c89f1aa : SUCCESS in 1h 06m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/665d02e414da4f458c8ce767950e1d49 : SUCCESS in 1h 29m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cc7c2a3b844459182c16d364a7b50e6 : SUCCESS in 41m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b1bc9831c3a4980b362738796fb4d9a : SUCCESS in 17m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ef0da104aef4daf88393a9d84301f80 : SUCCESS in 27m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c4b891c1b313490ea563231ed1d7fb89 : SUCCESS in 3m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69896d087522441fab00a0a04a97bd93 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a169db86392c499c90f33277fed8f006 : SUCCESS in 5m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5f8c5229582049e78125547e35e10ea6 : SUCCESS in 9m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e687090d8754a57828273ac46383264 : FAILURE in 21m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e88041fabecb4eb6a27390ec0abaef08 : SUCCESS in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15dbd106404b47a9b8e5881fb652e159 : SUCCESS in 1h 52m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/abc6e39404cf47d2b46f8521e1f8cd48 : FAILURE in 2h 14m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb0e873a41a84f8985415e8bd7201212 : SUCCESS in 1h 38m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7336ccc4783047c1af3999175062d2b6 : FAILURE in 19m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fb4dbaeaaf644bfe853f65bf3bf0ef70 : SUCCESS in 29m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b96c7a3f974842099ee5a83a9f4613df : SUCCESS in 1h 11m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9ebd3c8c50554ed39b1d01e7d7bcbccf : SUCCESS in 2h 01m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a264bb7202d14b629d0a7980ca0f2c30 : SUCCESS in 25m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/83700c75746949fb9670a8410fb35547 : SUCCESS in 2h 01m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b82493a1ccce4746a5f336e074a1c011 : FAILURE in 11m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c6f586ffa461415e81a3ce7839fb55b9 : SUCCESS in 53m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ab36b181654730a56a61d86ef58ec6 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8349737f3910a8830b59493713029bb75041f285","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-18 05:00:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-fc : SUCCESS in 1h 3m 59s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-iscsi : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4f7fed1a8c323f7b8ed263f6cba1fdd029c2d8c7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-18 05:55:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e5de9b47408af6a07d686fffacc80a215814637","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-18 06:00:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe20397df6871eeedfba1a2fbea17a7ed127c481","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-18 06:17:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9aa3a22b2472e10cb7b0040176cb08d97d25a4ef","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-18 06:23:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ee8fb18c2d2694bdff6624d3c0cab64f9d0b2a5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-18 09:29:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-fc : SUCCESS in 54m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-iscsi : SUCCESS in 1h 14m 51s","accounts_in_message":[],"_revision_number":2},{"id":"a5c8a56b998b59659281600bf26e31b8edefd4ae","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-11-18 12:32:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-fc : SUCCESS in 59m 40s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-iscsi : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e8efae3e5850b8f48bf7583f389e0f9328264f6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:15:53.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2494/.","accounts_in_message":[],"_revision_number":2},{"id":"8495c2d72d1ceaf3a4b661cf249bfdb18516acf0","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-21 01:34:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"812eb16a9ceb96d8a386c3746e2bf595220e17a0","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-12-01 21:57:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a800a3b3a052e731de22e9ebf24ac1d12fd6725","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-12-02 03:40:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e23221f34b7dfaed29919b8a8b0651764e2c93b","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-12-19 09:45:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e89fdb347a27aefc928d1aa03912e8a4031df60","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-01-02 16:35:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c095a3fd4c8b7aa1068b493b0a0f44e66f80aa1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:36:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd54e2a9a3e0dcb4753f0400c037bfbee78f1600","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:48:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c7d5ec95df43cdac38d39cad18e6c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384ceb0a36d9477eb006a31fdb5620bc : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e709bf907a9f4173b9f0a5c73cf4e999 : SUCCESS in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdfd0994981a41ef97126313aaf258cd : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519224ee7eb14fefafae2ff954168277 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/65521cb9140d4bd5b85c8e203d91eebf : SUCCESS in 1h 01m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/730609fb8df541578de96e79ed7e56a9 : SUCCESS in 1h 07m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49a62bd7c37a494b93c92c1431c3dd79 : SUCCESS in 1h 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f6a55511a9b4207bd227a7ab80e520b : SUCCESS in 1h 00m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0e5faf2ffb242ba8b5bba06bb3fcc7f : SUCCESS in 11m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5371f47255b9481f8647a3d49fc5c985 : SUCCESS in 1h 08m 21s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/397c9e5b289c4222afbeaeba5bae65c0 : SUCCESS in 2h 01m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59321f1631f5472e881590d6d3cdae00 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":2},{"id":"cd0e8dff980d24f05cd082a09eb69c8676f4037b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:48:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"203716218fe7c14063dc1721579892a3c090f091","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:49:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4d415f2acb4cbdac273c257263eab1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a934710f453476fb9656e6f1350f63f : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d433f372d1cc4212a9dac8eb5ef872c0 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"96e4fe1d3c7c78d2d9147c524e14eae6671e1543","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-01-02 19:35:58.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/45/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"66236283cf315e631ea3c39b802cdbd0a1b5241d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-02 23:27:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-fc : SUCCESS in 1h 3m 19s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-967146-2-iscsi : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"606dc17cd35e4ba0d14edbc796cfdb31d2ffc14f","revisions":{"516e37684d4f5dde8439cdb52786969e1aa20caf":{"kind":"REWORK","_number":1,"created":"2025-11-14 01:49:35.000000000","uploader":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"ref":"refs/changes/46/967146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/967146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/967146/1"}}},"commit":{"parents":[{"commit":"1d59848eb6ea960cdd52c0698c84c251b0c4b788","subject":"Show volume attachment host information for services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d59848eb6ea960cdd52c0698c84c251b0c4b788"}]}],"author":{"name":"anthony gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-06-02 18:20:46.000000000","tz":-420},"committer":{"name":"sapd","email":"saphi070@gmail.com","date":"2025-11-14 01:49:19.000000000","tz":420},"subject":"Hitachi: Add support for Hitachi VSP One B20","message":"Hitachi: Add support for Hitachi VSP One B20\n\nChange-Id: Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\nSigned-off-by: sapd \u003csaphi070@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/516e37684d4f5dde8439cdb52786969e1aa20caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/516e37684d4f5dde8439cdb52786969e1aa20caf"}]},"branch":"refs/heads/stable/2025.2"},"606dc17cd35e4ba0d14edbc796cfdb31d2ffc14f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-16 02:44:49.000000000","uploader":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"ref":"refs/changes/46/967146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/967146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/967146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/967146/2"}}},"commit":{"parents":[{"commit":"1d59848eb6ea960cdd52c0698c84c251b0c4b788","subject":"Show volume attachment host information for services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d59848eb6ea960cdd52c0698c84c251b0c4b788"}]}],"author":{"name":"anthony gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-06-02 18:20:46.000000000","tz":-420},"committer":{"name":"sapd","email":"saphi070@gmail.com","date":"2025-11-16 02:44:14.000000000","tz":420},"subject":"Hitachi: Add support for Hitachi VSP One B20","message":"Hitachi: Add support for Hitachi VSP One B20\n\nChange-Id: Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14\nSigned-off-by: Sa Pham \u003csaphi070@gmail.com\u003e\n(cherry picked from commit bc5591a1f9ceba64eaf5e5200b0bed60f79914ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/606dc17cd35e4ba0d14edbc796cfdb31d2ffc14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/606dc17cd35e4ba0d14edbc796cfdb31d2ffc14f"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
