)]}'
{"id":"openstack%2Fcinder~967158","triplet_id":"openstack%2Fcinder~master~Ibfc8b6248f28487929f72df59f4dc84257ed38cb","project":"openstack/cinder","branch":"master","attention_set":{"38590":{"account":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"last_update":"2026-07-08 06:11:56.000000000","reason":"Someone else replied on the change"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2025-12-04 05:57:22.000000000","reason":"\u003cGERRIT_ACCOUNT_38590\u003e replied on the change","reason_account":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibfc8b6248f28487929f72df59f4dc84257ed38cb","subject":"HPE 3par - support for autoSynchronous as part of RCG creation","status":"NEW","created":"2025-11-14 05:43:00.000000000","updated":"2026-07-17 15:07:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5083963af2ca94b782fcf3d67a987ef68364e5bb","_number":967158,"virtual_id_number":967158,"owner":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-08 08:23:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":1,"date":"2026-07-08 05:57:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":1,"date":"2026-07-15 08:34:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"}],"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 05:43:50.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 06:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 06:44:04.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-14 07:47:38.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-14 08:40:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-14 08:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-14 09:22:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-21 14:40:42.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"CC"},{"updated":"2025-12-12 12:06:32.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"CC"},{"updated":"2025-12-12 12:06:37.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-07-08 06:11:56.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-08 07:16:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-08 08:23:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-14 10:11:12.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-15 08:34:10.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"f46a3f7f000a03b45b30820b816862114188dfc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"date":"2025-11-14 05:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2636bd0a1804ace185ccff049e1ca21ebfdfab1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-14 05:43:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"db694f82da5640029c2b492e0a3d59cdde8df0da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 06:08:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9d32ca81e440fcb7d9156b9a010482\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c282184e0287482ab530391f1c1e7183 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c299104a9198419890cb8aafe8ec1d65 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d31c64c7c95546d6b9b41335de3badc7 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f4b7246f7397ed34543840685381210da96f49c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-14 06:44:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibfc8b6248f28487929f72df59f4dc84257ed38cb .","accounts_in_message":[],"_revision_number":1},{"id":"3f8b2020449560d16e316a36231f39ece22d7088","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 07:47:38.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2599/.","accounts_in_message":[],"_revision_number":1},{"id":"4ad53c4d4e7dd52c7f319ff50b3b2352f6ddb774","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-14 07:56:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2485/.","accounts_in_message":[],"_revision_number":1},{"id":"50fc8fdc50f9275b6a2f31da66cd01a7737e0d76","author":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"date":"2025-11-14 08:30:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcaa6c304be92099e25f9c143f10bad54da4e3e1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-14 08:40:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/967158 : FAILURE in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fdd9e317127a57417af701d277bc1478ea5f576c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 08:44:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03cfcec242f46288597aa3ccf5af811\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d95f5cd17d943e8ba1f864184329698 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83b78ad73cca4cdfa32ec5bbf13b9cb1 : SUCCESS in 8m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb3819fe3e5649ce9c77af8e8ff1a852 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39619cf286740bf82e6ad3be28d3948 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/da210da2ee2a4a3f8916e0408fdd3da4 : SUCCESS in 39m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56c98ad6b1a14c5fafe6fc5c111c092f : SUCCESS in 35m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5832e11a0334ba59c6881f17b979c00 : SUCCESS in 1h 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44fd944a96d742f2ac24a4c16976ec29 : SUCCESS in 32m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b9d4cf205d3540af9db9fb9207f5ae13 : SUCCESS in 19m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a71e7e4dd6048c7815329f38117508e : SUCCESS in 6m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/005b60528ee14aada0dcf97575ce4c60 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ff9135e44bb4e23a977dfd56d718251 : SUCCESS in 6m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3e1a943e5a174d9dad30dfa312fca212 : SUCCESS in 8m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9c0d7bcefd84e79bc6a79cbc0d0ea21 : FAILURE in 21m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5e6df1c6fc4d4582ac00bdb0f32507dc : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5e3ed2914244f46a4d8f638e54ad903 : SUCCESS in 1h 50m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce00ffdb62634c7aae17bd9ff3391c68 : FAILURE in 2h 52m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/54d04f909c754e73acfc6bb60c4a122b : SUCCESS in 1h 55m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ce8d7574b488421bb433073c5a7d5309 : SUCCESS in 1h 33m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/98a4da0ad55f4f9c86a8a21bdd697130 : SUCCESS in 36m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/93b0308fb3744b3d81d31e2bd3e374a9 : SUCCESS in 1h 07m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d47a4d664e184705af9f705265f7423d : FAILURE in 1h 45m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/304e41fde93048bb9c0d9f2c2d8fde5e : SUCCESS in 45m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/270d606efd7e4cd8a444ef3c24e4d42b : SUCCESS in 1h 38m 00s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/046789249952463fbd3c37c54de7f2f2 : SUCCESS in 1h 29m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ebac329cb4a47cf84138f41cf5bc5ad : SUCCESS in 1h 39m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da53cb46743c4f789e5d7223253a51f4 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2b1cf9fcd3ce099a9bcec115e598046e1bef8a9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 08:49:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb488695e7142a89b8e35ab7b88e69f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21af75f1251f426d90f716557d947cf7 : SUCCESS in 15m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd949c81af574d489308ae40156689e0 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37e8d9f4e4504c7ba77ae77a022209da : SUCCESS in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34aa29ca99b9ac2d91ac3245946b8c959b4becb8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-14 09:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/87a06a67b80f46c78b39a81f766b737a : SUCCESS in 43m 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3c69de77bfc741f38b9c9da9afb9edfc : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dc08c72d5e3e9461beb7158b79fb66648d6589ff","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-11-21 14:40:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b91d4c43939da47108d749b58ab196fde3487535","author":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"date":"2025-12-04 05:57:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abeba0d27fe736d82b2f4388379906975aa7e995","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-12-12 12:06:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"918224daadc10933fb9820719200d9e0dc6ac8a3","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-12-12 12:06:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"806d7c1086e472542fe379dd38d902d41b06c4b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"real_author":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"date":"2026-07-08 05:57:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_38590\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"}],"_revision_number":2},{"id":"7a158e86780dd77d9e4bfafee2cad97033c8fd7b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 06:11:56.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/530 : FAILURE in 0h 14m 35s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/967158","accounts_in_message":[],"_revision_number":2},{"id":"3616f05277cd5020b3ff3adf1d59937fb223eb8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 06:41:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509e2d80d6b74719b48734d1b84ecc8f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df6939182a74446abaa42efce0009a40 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed037141ff4244c2b352480b99c0275a : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/945c71c7d5b24b37a695186226c044e6 : SUCCESS in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0624877f497f1f950bd5e5bff3c82e6ede02743a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 06:44:51.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/802 : SUCCESS in 0h 47m 27s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/967158","accounts_in_message":[],"_revision_number":2},{"id":"a3b3afbc8de11a816be79e371e4a0e651d854dec","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 06:45:38.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/446 : SUCCESS in 0h 48m 12s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/967158","accounts_in_message":[],"_revision_number":2},{"id":"ead2b7c642040c16f28da0fcc995209cf48b9f43","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-08 06:59:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibfc8b6248f28487929f72df59f4dc84257ed38cb .","accounts_in_message":[],"_revision_number":2},{"id":"64f9a866c78d06511911488367c466671240b8fc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-08 07:16:00.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d01beda567374d2cb25e1c0f3c85c5f9/","accounts_in_message":[],"_revision_number":2},{"id":"994d0a481be365db522bfba7d02854e3cbf9b14e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-08 07:50:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d01beda567374d2cb25e1c0f3c85c5f9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/505365d1ac4e4fc99742eb1f17ca8414 : SUCCESS in 42m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/799fd5cb90ac4589b22d003613d14120 : SUCCESS in 40m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f69c5187ff648c6b7d4c0985a631899 : SUCCESS in 39m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a35ac6644d0242a78840870ab97b9fd2 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9098bee1f21e81f2161a59857cd915be7d34a08a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 07:56:34.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/531 : SUCCESS in 0h 45m 21s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/967158","accounts_in_message":[],"_revision_number":2},{"id":"81d9ba3da9efa9299a17c9b6d80f818b3642e0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 08:23:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a71d68f029e4e95bcdcaa92ca9596f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23939f5fb66a49e097563dea36c29ee8 : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53a13c2b442149238f9c04ac1d84c9af : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9d87e18052f493b9b7c1e2a559a462b : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4139ec2770db4b2194ef426e1d7e61fe : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585e1ef2c34241a6ad3d20bc7377a53a : SUCCESS in 16m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/4d94c7f4fefd4c10b2d7bd295ad3b53c : SUCCESS in 1h 11m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de79a7cbcbac474785cf7748b1983b76 : SUCCESS in 57m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/95984d2a4f7a4949a1ed82866531e076 : SUCCESS in 1h 49m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/361fc2822b9f4de5b99da9a1a06876bb : SUCCESS in 35m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d85fc86e967b40d3b3238e0df65754f7 : SUCCESS in 12m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0327989a7054e888a4d9f7558f25dae : FAILURE in 26m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ebbbc8d7a5f4ff2a9276b486432e736 : SUCCESS in 6m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dfa1c328ab7442ba9a97a49c3fc570b7 : SUCCESS in 3m 17s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/d414d05a8e514e07bc541fa0f8253742 : FAILURE in 13m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/47205304557b44e48064e60688768086 : SUCCESS in 6m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/795c7f9124fa47f4a666c37ca595da33 : SUCCESS in 4m 30s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c6d10a6b59bb46c3a7f002f1f9320996 : SUCCESS in 6m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2228463f797147e18fb126f060e9af80 : SUCCESS in 43m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/004982361d514223bd6b055a972fbaaa : SUCCESS in 6m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7935c004482b4a6e874c3b3489f3c139 : SUCCESS in 2h 12m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/25195daff25a422c8f0cd43398b6da5c : FAILURE in 1h 32m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c41fd27446514fee85c2aab23bb657d2 : SUCCESS in 2h 18m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ef137438bf894ebcb3f22fe9cbdf3f23 : FAILURE in 2h 25m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1c554b4e65514012bcaa0e3e76f237c5 : SUCCESS in 2h 25m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6ee3b410f76e473d899b0729726acadb : SUCCESS in 17m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/164d5f06e4de431b8f1f1b0c76c643e2 : SUCCESS in 1h 04m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ab74828e53343ad90ac943d5dca33b5 : SUCCESS in 1h 31m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7b716dbd70a648319d71f89bbb7c8384 : SUCCESS in 47m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ee38eb621294fc3a753c874b1e6d796 : SUCCESS in 54m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4b8472abecf459284fecce2da81711f : SUCCESS in 52m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"83ac183179667db68fdd86ec0533491191893e3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-08 08:30:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2680c27a5af441e5a2eba7c369667a73 : SUCCESS in 52m 48s\n- cinder-storpool-tempest-iscsi-multipath build/4ff79b8e304941a9aeb4f0b916f0f01c : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"150212f8226d7490555bd30f9eceb4e4caa319d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 08:33:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4722/.","accounts_in_message":[],"_revision_number":2},{"id":"be2fd529e9a7cab64aba9f2deca28875f4c26e44","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-08 08:35:34.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_07_2026_08_27_26\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"3546c88358b861cf268cbc0eb58d3b49f1da9776","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-14 10:11:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-967158-2-fc : SUCCESS in 59m 27s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-58-967158-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b1bb613b311e2830cb94ab88ddb6e8474176d838","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 08:34:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5083963af2ca94b782fcf3d67a987ef68364e5bb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 15:07:54.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4546/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e23f848179d21b6f9271be7698d86ddd36940f89","revisions":{"bca9c9ba3dcc127b2baf808c2dee4b5c5a91236f":{"kind":"REWORK","_number":1,"created":"2025-11-14 05:43:00.000000000","uploader":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"ref":"refs/changes/58/967158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/967158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/967158/1"}}},"commit":{"parents":[{"commit":"c27091e748657c02f181300dfaa7ed2141deff0e","subject":"Merge \"Handle volume_clear_* opts consistently\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c27091e748657c02f181300dfaa7ed2141deff0e"}]}],"author":{"name":"susmitap","email":"susmita.poddar@hpe.com","date":"2025-11-14 05:29:51.000000000","tz":0},"committer":{"name":"susmitap","email":"susmita.poddar@hpe.com","date":"2025-11-14 05:39:33.000000000","tz":0},"subject":"HPE 3par - support for autoSynchronous as part of RCG creation","message":"HPE 3par - support for autoSynchronous as part of RCG creation\n\nCurrently auto-sync is not happening due to which we need to apply failback steps manually on primary array once primary array is available post failover.\nHence applying auto-sync option during RCG creation.\n\nChange-Id: Ibfc8b6248f28487929f72df59f4dc84257ed38cb\nSigned-off-by: susmitap \u003csusmita.poddar@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca9c9ba3dcc127b2baf808c2dee4b5c5a91236f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bca9c9ba3dcc127b2baf808c2dee4b5c5a91236f"}]},"branch":"refs/heads/master"},"e23f848179d21b6f9271be7698d86ddd36940f89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-08 05:57:00.000000000","uploader":{"_account_id":38590,"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","username":"susmitap"},"real_uploader":{"_account_id":38274,"name":"Jyotsna Lothe","display_name":"Jyotsna Lothe","email":"jyotsna.lothe@hpe.com","username":"jlothe"},"ref":"refs/changes/58/967158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/58/967158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/58/967158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/58/967158/2"}}},"commit":{"parents":[{"commit":"7db0225cc8e08477e56cb4e82fc2666033bbc2ae","subject":"Merge \"Add agentic coding guidance and docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7db0225cc8e08477e56cb4e82fc2666033bbc2ae"}]}],"author":{"name":"susmitap","email":"susmita.poddar@hpe.com","date":"2025-11-14 05:29:51.000000000","tz":0},"committer":{"name":"Susmita Poddar","email":"susmita.poddar@hpe.com","date":"2026-07-08 05:57:00.000000000","tz":0},"subject":"HPE 3par - support for autoSynchronous as part of RCG creation","message":"HPE 3par - support for autoSynchronous as part of RCG creation\n\nCurrently auto-sync is not happening due to which we need to apply failback steps manually on primary array once primary array is available post failover.\nHence applying auto-sync option during RCG creation.\n\nChange-Id: Ibfc8b6248f28487929f72df59f4dc84257ed38cb\nSigned-off-by: susmitap \u003csusmita.poddar@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23f848179d21b6f9271be7698d86ddd36940f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23f848179d21b6f9271be7698d86ddd36940f89"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c27091e748657c02f181300dfaa7ed2141deff0e","ours":"bca9c9ba3dcc127b2baf808c2dee4b5c5a91236f","theirs":"7db0225cc8e08477e56cb4e82fc2666033bbc2ae","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
