)]}'
{"id":"openstack%2Fcinder~967260","triplet_id":"openstack%2Fcinder~stable%2F2025.1~Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14","project":"openstack/cinder","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"27488":{"account":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"last_update":"2026-03-18 19:24:28.000000000","reason":"Change was submitted"},"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-02-11 14:49:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-01-06 02:42:27.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-01-13 14:46:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-03-18 19:24:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14","subject":"Hitachi: Add support for Hitachi VSP One B20","status":"MERGED","created":"2025-11-16 03:00:30.000000000","updated":"2026-03-19 00:12:19.000000000","submitted":"2026-03-18 19:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967260","meta_rev_id":"d4fda4a6e5bd8366046e6595d48a94fe53f0659e","_number":967260,"virtual_id_number":967260,"owner":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 19:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"all":[{"value":1,"date":"2026-01-06 01:27:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":2,"date":"2026-02-11 14:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2026-03-18 04:56:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-01-13 14:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":1,"date":"2026-02-11 14:49:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-16 03:00:30.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"state":"CC"},{"updated":"2025-11-16 03:01:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-16 03:26:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-16 03:56:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-16 04:00:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-16 05:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-19 05:16:00.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-01-05 12:42:25.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-05 13:51:01.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-06 01:27:03.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2026-01-06 01:27:03.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-01-06 01:27:03.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-01-06 01:27:03.000000000","updated_by":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-03-19 00:12:19.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"87f155ab3e6015ec695ebdae99b8c50408e44268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2025-11-16 03:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b780f84f8ba65c2060a96fa5e3832e4df835c84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-11-16 03:01:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm 3par-iscsi-driver-master-client-pip-ssa02-dsvm : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e798fccac5078744532d69d28b1e27adda7a76f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-16 03:26:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b56307aaa0643d681aa4becc4001937\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eae909eefc334885ad1ce2cda5774342 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e82e56cdd8644889b172191c9bb778fd : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f6397af1b90bb58e402689d3b3ae2489c313923","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-16 03:56:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14 .","accounts_in_message":[],"_revision_number":1},{"id":"b3a38e90c3809763ef5a14950d643ab227e6e3de","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-16 04:00:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/967260 : FAILURE in 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"81b700b68c082835ec27376d2146881cdf65eb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-16 05:47:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85c10415a1f243c39cdf396070578532\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67568e8c5e9d4156af1da126a729dc34 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857679d40d174720bb1346a31faf803c : SUCCESS in 9m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/442c4d0e329d483b89fe53b5155e377a : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d06ef14e8984c63a899c0f81fb438f9 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/06980ed915804d3ea7c6ee20d7c9b1e2 : SUCCESS in 46m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2221a95939c3414fb1e5debf67ebfb7e : SUCCESS in 32m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3cb3ba29d084d8fb8dcd129c5f01131 : SUCCESS in 1h 31m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57f4489c2a7f4399a5621f80638c2e0a : SUCCESS in 56m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ffc4830591145458d3af80056cc4fbe : SUCCESS in 15m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c181a4b69964412a78f7fb18f410b83 : SUCCESS in 21m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0d3503b7ba2c4b04b098f3a82381fc64 : SUCCESS in 7m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8a9938cd004411ab542bb92862aa1b5 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2272328574d94073ab3e46979cc0545c : SUCCESS in 9m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0c529108f8634cf5afd1845ae6bb31e6 : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f856ebdfd3c24f0d9bb38ab76c08c44b : FAILURE in 21m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f254e246f8134b3ab9258f10de938a44 : SUCCESS in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20f0232bc6ca4304b4291bca5d21acf9 : SUCCESS in 2h 06m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1ad3b25d2184476fb49b62d459d56559 : SUCCESS in 2h 31m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/658a07c677f24009a4122f1ed9450ef8 : SUCCESS in 1h 50m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bdab6f14d10e4bdea03ebe995c4a0395 : FAILURE in 30m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5a4e0033bce648e099b17a4e379fb60f : SUCCESS in 33m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3cee22e96e8c46089b5adb0eecba9975 : SUCCESS in 1h 09m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3870737e712a46cd881132d7080e6c34 : SUCCESS in 1h 18m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/efaef24b146746cf9226bbc499f97670 : SUCCESS in 27m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/438e250bfef74dcd99cd0de1454d0ade : POST_FAILURE in 2h 42m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/55fbb2e6215a4ec898e40f162cb21b10 : FAILURE in 19m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37a98bfde43c41deb26fae7c8061e73a : SUCCESS in 1h 11m 32s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af66fc31cdf44dddb68a0b9e368611a4 : SUCCESS in 1h 39m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48b779477b4c48b3951577f4ad1d57a1 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e22527aec279ee91955ec5b0a48ac26f4bd0b1d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-19 05:16:00.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2495/.","accounts_in_message":[],"_revision_number":1},{"id":"7a8bcd61b588dff24722d3e7c19d78db85a5b89b","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2026-01-03 05:10:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c320cbced49b046c02f255e5fdee31bbccf06314","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2026-01-05 09:52:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87259ea36d6825d1447b2ed9d379297df6e79ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 12:35:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bafa6cd5545456f8084362e3364adef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a22d97d87743c5bd7de8e2b2d277e3 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f4a9ecfafe4214a83a733fabec4cce : SUCCESS in 13m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8171e8cbda8457c957a372e760ab58a : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c25ee1dbd04b5cace5643451bf8181 : SUCCESS in 16m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/7601f5ee812a4f458df5a6a59e721df5 : SUCCESS in 1h 03m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd3252670ea04d20b6ce8a151d3a6d7c : SUCCESS in 55m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b565d05dd414467be3df4c049aa1201 : SUCCESS in 1h 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f95a2d628c54c93b88a996bd000da33 : SUCCESS in 1h 05m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6e8e7810ad1400287ab38b822fcc800 : SUCCESS in 11m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/45dae2a760a442c1914dc94bac526892 : TIMED_OUT in 40m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/47a83cdce1d545158581f733b99ce65d : SUCCESS in 6m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4f1c3f52e1804a49817b4797b4595495 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5023b51d1ea3400b82a3953ea5402941 : SUCCESS in 10m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/09925c3883634eb9a819b76879193da2 : SUCCESS in 4m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11c51c88aedb40e389c022c67ed42f19 : FAILURE in 38m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ca513775f8f04f32986dcb9284fee52f : SUCCESS in 7m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72ede216e1874439a3432d1ecd43a2a7 : SUCCESS in 1h 06m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60276c0011bc4e41a1bd3b036a570fdf : FAILURE in 2h 33m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/880cfdad10064ef5a4902f1118ddfd96 : SUCCESS in 1h 43m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6b3a72e9863546e4b96ac4bba2b1afb6 : FAILURE in 36m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0a3309e582a849e69baccdee4d8f0fe9 : SUCCESS in 36m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba5bc852e471433f995923fa605c529c : SUCCESS in 56m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7f525d5b1ecb4cdabc3b39e44152e9d0 : SUCCESS in 1h 27m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a798818598494f429532d4c0a630250b : SUCCESS in 23m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c4d89b2d01644b9f89e0d41aa544781f : SUCCESS in 1h 12m 36s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/66cd52e6c372482aa6c3988d49879828 : FAILURE in 34m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0cbbff9ffe4b46a5a0d36c7a5fd4fbb3 : SUCCESS in 46m 22s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10c7d73cf6994d5caa9da295afc92370 : SUCCESS in 1h 21m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34b4e05ce6bb446fa3040fdd59ebf016 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"691e2ac8dda63dbba83031f9afc8d94c8089256b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-05 12:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_01_2026_12_35_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"48c737d769a8a72b8578606f4a38aca6c1eea717","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-05 13:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-60-967260-1-fc : SUCCESS in 52m 55s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-60-967260-1-iscsi : SUCCESS in 1h 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"94974745a40bcba551ab2ec4470fb25e76c228c0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-05 16:46:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-60-967260-1-fc : SUCCESS in 1h 9m 54s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-60-967260-1-iscsi : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"498676fcc0ded43c0093404fd5a9d4efc90b4be4","author":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"date":"2026-01-06 01:27:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55c80c522d27d99d32a8707c8c3db74d597e0a92","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-01-06 02:42:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c0adfc0dc6ac37c0ec4ea3874fc66f2cd463109","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-01-13 14:46:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8468c4525977aa6db354b39a0b83e40acc01ff9","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-11 12:46:33.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ded6d12ae9f578ec25728f67fcc9469929afd228","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-11 14:49:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de6035e9b878df087e3a62cdb405ceef4873f303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 14:50:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ccb6a03ba60cef513c7b29c002689d5fb865a9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 16:55:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0881a26873314967aa66f9f00e284ea4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77440ccde7e44ef99e5eaa2218af817c : FAILURE in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b14f6a5dbbe94c72adb9c17a2c8ab558 : SUCCESS in 12m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6273f84372748bfad4593215e09e8f9 : FAILURE in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d656c50cb744c04853eacbe47231751 : FAILURE in 4m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/d7b019af7d0b461e9a39cc005dde7afd : SUCCESS in 57m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79af4b55b2e94ac8833ab96a8c748f85 : FAILURE in 44m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53ed8c61e78540fd93c321b632c65c19 : SUCCESS in 1h 30m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58e856e6fe8f474682ff683bfe0c4f51 : SUCCESS in 38m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e4d15623cf94adfaffe55e2ec0660ab : SUCCESS in 11m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3e3c3c1f402840089e84af9ac2ebf4cf : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6f0a3545b452458b84b49cbe51a999c4 : SUCCESS in 1h 59m 15s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6beaca3276734b508d6c5cbb30db9f58 : SUCCESS in 1h 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3cbf5e872f4200ac76b77df7e9dc14 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fbac8d822ed0986e43b6eda3a88b52fddf921d66","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-11 17:07:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3f23c645dbdaf2f9eb4004c5e19f646f9006bb1","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-11 17:51:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17261200700e6d10f3a13ba8a935da84bc016937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:32:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2570c87ef9584a55a2c85e31c234bf0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c592225e604ef5969f3a11e6513195 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1cbf5b882f24f96923c97cbaea0b9bc : SUCCESS in 12m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63297fa4f30e4fd3a8717a21473be8d9 : FAILURE in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10be259fa43344869f160185386e6dcb : FAILURE in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f194b48f0f9641e09cb1fa1d1304ca92 : SUCCESS in 59m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7a68057d921246f9bc3b654b69b36ee2 : FAILURE in 26m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/127767220659487790919ee41fff1c56 : SUCCESS in 1h 13m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e72a18c3e0a44f9a36c9cdc3ed08574 : SUCCESS in 1h 01m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e499bed69094a88bf66e6ef45b5d09c : SUCCESS in 17m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dcaaf77e1fda44d89cdccf2a534f759d : FAILURE in 3m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c0d3f0b420104a938e0d384032c7a553 : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8beadfd1b4bd48358aecccd8973b315c : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9dc2c87093947528f068c76eb3c09a8 : SUCCESS in 6m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/272c4ce4fba1425689cf19baf28cb43a : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cea80078cc404b00b16f8972c5527527 : FAILURE in 34m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f56754b3206f4a59970d8c87c32dbbf7 : SUCCESS in 3m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7716bf7f35c04d3ca34d8ca46c2e5f53 : SUCCESS in 1h 25m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34bf8b70e75d4f64a9332a1852b55520 : FAILURE in 2h 20m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7cb6e099ea54f3c80da2141fde1cf38 : SUCCESS in 1h 43m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/03e8635e9c8d4a0db6356bb32492fb0e : FAILURE in 24m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2cba486709cd4cf0a3967ea68789b571 : SUCCESS in 39m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7388e92e3f024606af44f8e73d967ef3 : SUCCESS in 55m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e579b3fb90954cbb981c6f905579994b : SUCCESS in 1h 51m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ebeb27e755644769aa173c2614d79283 : SUCCESS in 22m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb6b69d39ca448f6900da22ca1f52ba1 : SUCCESS in 1h 57m 33s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/22c9ac5405c044f5ad743d0ffa5ac0d4 : FAILURE in 26m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4719d3856e584219874cedbc019d59b8 : SUCCESS in 1h 43m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e27b16bda25e42ac8693bd6915f3fbe5 : SUCCESS in 1h 31m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8ef3dd01194b0e8d953cfaf054822e : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":1},{"id":"67b64813a344ddf09b9cef3fd2c7a05fb9b470d2","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-23 23:50:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e8d98c7341ae5aeed7c5493e87f25ef12b47215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 02:00:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f31e70210604fec90a32cdcf675c6eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fe889557c64bc9a75a5d67b81a2b9c : FAILURE in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/883baa5455b84fe3995bc239679887f2 : FAILURE in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f47d8f98c44043ce972e06a0432ef303 : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a33597ccb24f1f98e0fa86c52bf9b4 : FAILURE in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/bd846f5a0e864e018ced8f210ee9a7bf : SUCCESS in 1h 01m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cedcec1b81c4e7cb756498425a623c3 : FAILURE in 47m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/038a83ddfb754bb0b18fbf94eb882c87 : SUCCESS in 1h 30m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bee28a631244d889a7d1063c179a703 : SUCCESS in 1h 00m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8adafeb1ecf4f2584b625802cbcdb04 : SUCCESS in 12m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4da87844e85d413ba3d4c967b9375de2 : FAILURE in 5m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a379760bdb964aa984151e253a35047a : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/402c4d0d208b4bceaf6a11d08ece05c6 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aace230d85444dc4a7daffabd4ec48e8 : SUCCESS in 6m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68dc2c12cd13495d98d82f0d13d42eac : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5af38c93b60a47239c460d47b65ab5a6 : FAILURE in 36m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b0af3180217421c9493ad7935e295ce : SUCCESS in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0347ad219d524d36a1201d0dd80150a8 : SUCCESS in 2h 09m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/341463e7a6aa41e389b76088fbc79b9e : FAILURE in 1h 29m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d5c6180134a4000b433374ebd6b8a95 : SUCCESS in 1h 35m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ac96cd9104b24424af364d2414137107 : FAILURE in 15m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aa2a5f7a95104be39842428d558d6de8 : SUCCESS in 17m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c5ef59333870441e9cde5c9232682852 : SUCCESS in 53m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7908c973882941f38380d64a8d6495b1 : SUCCESS in 1h 18m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6a6d2f5d38774323950e4ad4250e0e51 : SUCCESS in 30m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f7f182fbc3548a28db5d5607711046c : SUCCESS in 1h 04m 55s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f1bb3b5ade0f4494859de8edd1796557 : FAILURE in 23m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d191e7a1bbe246249f73cf010e0eb093 : SUCCESS in 1h 39m 24s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fb784368a4e48cb88ff721da476a5ed : SUCCESS in 1h 10m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/929b622e17294ce1aee994b4e41db4d2 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e5c25770d4647b616b52e7dd254242e3abf11652","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-24 04:41:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e13d88e2b91834ce97d3bb62c982c5a46dae261","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-18 04:56:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f55d9e66e894a9f6c716db523fcad53c5a83d1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 07:28:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bbf2d4f419e454e96e99594977ad209\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53b61def3aa42179059314438cea1ab : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fd6d6cdf32a45259f60e0b08b5c4c4d : SUCCESS in 10m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/526a0f8dbd864b5eb03cd87ed875a3ca : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dbe836f3a9f4bbd9593375a059d61ce : SUCCESS in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/627fd00dca4d4026a5dfce83196bfd20 : SUCCESS in 40m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f828fe6faaf14532b1ae099512b64939 : FAILURE in 23m 20s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3551723f24c4741acc243e2c10dc779 : SUCCESS in 1h 29m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a44699f689d6438aa731a1031792c07b : SUCCESS in 37m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2cb54dfcd2f4623a16d507a724ea912 : SUCCESS in 15m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/35465aae977b415c9bb53596356cd210 : SUCCESS in 19m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22dad319c7e443d187d2820a97a79c8d : SUCCESS in 2m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93e3daa5103e4c9aa66c77d6a4f30705 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6198d6138f194144947fa183861653b6 : SUCCESS in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/193d57eee4e34e4288d4c31f3a3d3e15 : SUCCESS in 3m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b8ade1f3385d41d293b146253d00d907 : FAILURE in 24m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e78a1e7ef6b144c09e0925e3949bd856 : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14c93d7b6ec849bc904b533135582528 : SUCCESS in 1h 19m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4e8fcb3f2e9746489d3cdedf1c9f5106 : SUCCESS in 2h 26m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08fa31dbc46b41aab626c596c895b9ff : SUCCESS in 1h 45m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/15796795d1684f6d8e0b766debe7166f : FAILURE in 16m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e39267e345a4570998a4da9b9973054 : SUCCESS in 25m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0571b61efa324cf492b8d5ace32163ed : SUCCESS in 47m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8432831664824e1887e66be217cdf69e : SUCCESS in 1h 12m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/221b66e3ea0b49868145dfe69794cd37 : SUCCESS in 43m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d58482dedeb0462686a9d0fbb0674b91 : SUCCESS in 56m 53s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8141ddce3cb54078943e66350170cf19 : FAILURE in 38m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e87149856644055801c9ed905d2bf1d : SUCCESS in 50m 20s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a52375035e44e1da8fd51cfb87e2143 : SUCCESS in 1h 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6642624c4af34bcb87b7880ca0d81adb : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9f53af4dcd156689e23d193104dbc5fb8153474f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 07:28:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d42329971c72132bc387dcbf22c38668c1dc4cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 09:02:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dfa6eb593bc4d539651bb6dbd7ea4c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0712785cca6f4937a7035cc4c6da520e : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0497e830239b409984963854fe111f1b : SUCCESS in 14m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0fd02f5104e24b76bc700a059bffccd9 : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec48bbdc0a644da849c304ae6f37ca2 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/5f30e42c1b3d49e2b4ab7d4801434a77 : SUCCESS in 42m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5693585e2a9d4c29a09c3dc3a3a1812c : FAILURE in 38m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b710a41a94e64e3d934d7faf3fa39db1 : SUCCESS in 1h 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0159ea3bf7c47f0b8eec5b7818125a0 : SUCCESS in 32m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41e52ff0b69d487a8ba53119ab23360a : SUCCESS in 19m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abced5c7aaac435d915aa84184137021 : SUCCESS in 1h 12m 11s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db62c15baf9f4ad1bba5a50f39a86a49 : SUCCESS in 1h 33m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d97cc3b6e004488f8ba19efcd11dab8d : FAILURE in 1h 00m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb7054c4bc146f88af7c73bec55b839 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9e1d431b245886b7324fd1d15ab7e18bf8a6fe29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-18 14:52:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a2ee220e1a98d7716db52c3d76740d241e97c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84cf080898784949990ba8c6871d08e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59036ac722c04c769d553e76a6b6286a : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc2eaeee743e481e876ebd20cd45e404 : SUCCESS in 8m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74a4c8044c664dd9af041847827146a3 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f96929cce649608e93cb952e7ffc21 : SUCCESS in 17m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/60a535aed2134c3f8b27b2a6f1cb74e9 : SUCCESS in 56m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84270453ddce4c32b280af1dd870d804 : FAILURE in 37m 05s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42de2d767d7048e6ac5514692c321453 : SUCCESS in 1h 39m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/046c788219dc4c48ace674b7ed4d1478 : SUCCESS in 56m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b219e41479ee4942ad5f3654bd4c1471 : SUCCESS in 13m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4b39e5df52934987b86ebef66ae1a9bc : TIMED_OUT in 41m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c8e22e77a0c84abaa5c31e7b6ec8adb4 : SUCCESS in 6m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dd26f813da443719f8330f375b93d94 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cde53c0415b647438bd48387f491db4c : SUCCESS in 6m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/580db6196e0846f6b553715759c95c60 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4e199519279e4a4d96f324eea4fa7ffb : FAILURE in 31m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0b94cca55c44ad9bf49bbfebd909c11 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc865b2bd79343cea61a08e57ac8085f : SUCCESS in 1h 45m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e219871007544c1f95228e3a10cc3e8f : FAILURE in 2h 42m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e46d4cd51173492c81a0385c61e0be2a : SUCCESS in 1h 49m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/389795a633174105b18d2b9fb2399d90 : FAILURE in 18m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aec7a7c3fb6b4d4bb32722ee2271ef40 : SUCCESS in 38m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ded81aa7b4ab490a8d1b46f5f9938516 : SUCCESS in 37m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/661db4b985ae45a8bde27d3aeb8b08fc : SUCCESS in 1h 28m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/abe41448356a4f81ac09d2c2d983f42a : SUCCESS in 41m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7439a97c8c0849f9a3bca20554b28dbc : SUCCESS in 2h 02m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/f42d7c03a0204f6ca858098438b8359f : FAILURE in 36m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7c6050d4e5a244bfaa6b3f5b855024f3 : SUCCESS in 1h 34m 27s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/567d9f5894124f6687ec521ba9d2e41e : SUCCESS in 1h 42m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7749979a958a419785b5d2f24191c1c9 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"63c687356bddd930b8b2213aa7be9d5870e63130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:41:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb88f3e5b44ee8798159a041de8968edd4dd7a47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:24:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60a79082ac14fa392a71fe6b87d6eea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbfa36a17954eae94418d91baa2659e : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d7a4a2fe5df4262aa5b7f3fbc1f911c : SUCCESS in 14m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7e50a055c694aedb0b1a137deac0c22 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e34c79f40fe408ebd1b622e65601e9f : SUCCESS in 17m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/3783b5e54d1a47e79336d1ebb4f031ac : SUCCESS in 50m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9336467b06b548e1894e5a3b0f08e016 : FAILURE in 23m 11s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/215751682d544949ac51e6c06fb65727 : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce814abf075046fe9eeb2efd3593f35f : SUCCESS in 1h 03m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/662f11f03c0b453c9647654a863d33c1 : SUCCESS in 13m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9132101a2f124407b4dbd3d7b676ee07 : SUCCESS in 57m 02s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f7d1ebcb51443e080e819bffdb716b9 : SUCCESS in 1h 07m 52s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a47d1e95b964a5e841a030946cb5b6f : SUCCESS in 1h 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b33060d3be7144ffb6d17a363f473f62 : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c3088b61d9e4379080971bb302b4f5b3ab36e0b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:24:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a1b14237425d88f2efd34b83a83eba69d1ea169","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:25:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c50569e34a4729a2f1a634883ba723\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd423b467c0485a9a486e0103d3b622 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4fcd83f1229487e986a3f52c1bdf208 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"d4fda4a6e5bd8366046e6595d48a94fe53f0659e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-19 00:12:19.000000000","message":"Patch Set 1:\n\nNo Builds Executed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1173/ : NOT_BUILT (skipped)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac1ee9bb8eba4d03d4594ee91a540a65cd9b4ad9","revisions":{"ac1ee9bb8eba4d03d4594ee91a540a65cd9b4ad9":{"kind":"REWORK","_number":1,"created":"2025-11-16 03:00:30.000000000","uploader":{"_account_id":27488,"name":"sapd","email":"saphi070@gmail.com","username":"sapd"},"ref":"refs/changes/60/967260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/60/967260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/967260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/967260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/60/967260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/60/967260/1"}}},"commit":{"parents":[{"commit":"0c0c5bf5ce1a2104eef20a5504b37d09b750e3a0","subject":"Merge \"[NetApp] Enabling total_volumes capability support\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0c0c5bf5ce1a2104eef20a5504b37d09b750e3a0"}]}],"author":{"name":"anthony gamboa","email":"anthony.gamboa@hitachivantara.com","date":"2025-06-02 18:20:46.000000000","tz":-420},"committer":{"name":"sapd","email":"saphi070@gmail.com","date":"2025-11-16 03:00:09.000000000","tz":420},"subject":"Hitachi: Add support for Hitachi VSP One B20","message":"Hitachi: Add support for Hitachi VSP One B20\n\nChange-Id: Ica909518b9a0e4e58ad757bc9373ffc5e52e5a14\nSigned-off-by: Sa Pham \u003csaphi070@gmail.com\u003e\n(cherry picked from commit bc5591a1f9ceba64eaf5e5200b0bed60f79914ab)\n(cherry picked from commit 606dc17cd35e4ba0d14edbc796cfdb31d2ffc14f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac1ee9bb8eba4d03d4594ee91a540a65cd9b4ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac1ee9bb8eba4d03d4594ee91a540a65cd9b4ad9"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
