)]}'
{"id":"openstack%2Fcinder~968380","triplet_id":"openstack%2Fcinder~master~I2c5df478acd1f5840ad8d572904144448e127f46","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-09-09 05:17:45.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-05 01:42:25.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-09-09 06:54:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c5df478acd1f5840ad8d572904144448e127f46","subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","status":"MERGED","created":"2025-11-25 19:39:03.000000000","updated":"2026-09-09 06:54:14.000000000","submitted":"2026-09-09 06:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":100,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968380","meta_rev_id":"368848510d4ade7b53c9c77c85d66f8ad6e2a91d","_number":968380,"virtual_id_number":968380,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 06:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":2,"date":"2026-09-09 05:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-09-05 01:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":1,"date":"2026-09-09 05:17:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-25 19:39:15.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-11-25 19:39:38.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2025-11-25 19:39:38.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-25 20:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-25 20:32:16.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-25 20:39:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-25 20:45:34.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-25 20:48:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-25 22:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 13:16:03.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-06 15:20:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-06 15:30:15.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-07 03:29:35.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-07 05:23:40.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-01-07 05:23:40.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2026-01-07 05:23:40.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2026-02-11 04:52:30.000000000","updated_by":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"reviewer":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"state":"CC"},{"updated":"2026-02-11 07:55:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-03-02 07:20:44.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-04-08 18:47:36.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2026-04-13 15:46:22.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-13 16:06:21.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-04-17 15:40:13.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-06-09 05:02:15.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-06-09 05:25:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-09 05:40:58.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-19 19:57:34.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-07-01 18:06:21.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-04 01:26:16.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2026-07-08 15:29:48.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-11 13:57:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-16 13:11:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-19 09:34:01.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-19 09:36:19.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-28 14:26:13.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"222a2375708aa3deead3120cebdbcb355fe20845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-11-25 19:39:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d7bf4ce0fbc71984ee8b5d53fc95b05da67e692","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-25 19:39:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/70/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d38c5115cf57ddbb5ef01eb78d613eca6f118031","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 20:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ab701e9ab64ba3b3828e0c28d92437\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a0e604844b34e349b798ea58391a319 : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/040602089b6945f99b67857fbaacda3e : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4e9e5084f4b4eef92a2be270730939c : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e672275180dd748e6938b102d6d5a579ef76a86c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-25 20:32:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2576/.","accounts_in_message":[],"_revision_number":1},{"id":"aafc9dfbc07faa962c0d6ca97b5d09ea5632842f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-25 20:33:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2724/.","accounts_in_message":[],"_revision_number":1},{"id":"f1899877b72e1afbff2ab1cba33ad86d4fb5bab1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-25 20:39:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":1},{"id":"fc152f95f7dc73d7241d02df12d812341e341471","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-25 20:45:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b2a5d45233604fa380162698118cb52e : SUCCESS in 1h 02m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/621ff1ac19414036b3cc2a91ef9d80c1 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"20a667ae04c5962afa11ebe062681ba4c464dcc6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-25 20:48:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968380 : FAILURE in 1h 07m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ff8de747e3c589cf54cbc4f5f9359517f9c89189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 22:34:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8a4deb1e2e2499cadce16fa83ec55a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d9b05e25c94477a1aa354f834dbc53 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7482d3dee5cf4c4daf52906e20c800eb : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1879e7cf59ad41d2849bae0040edaeb6 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457031bd9cfe41e487c9c49a027e2ef0 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a88effadb4964abb98cb17d78fdf3d9f : SUCCESS in 1h 00m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/442057679b54449cbe38fd19edb57a01 : SUCCESS in 29m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54ec2d12a6e04d7bbd3a1ca80963210c : SUCCESS in 1h 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa9e4229a86b43c68c3bd7c55f2e7db8 : SUCCESS in 32m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a028348984541b09febc99f70ba38cb : SUCCESS in 14m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7c2027057cc2456881b5aee225db0727 : SUCCESS in 19m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3a3b8413a88440d3ba833afa0ed9057e : SUCCESS in 3m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76e540cc870c41f295455d8bba0b9be4 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef01507c169e443da3e993c8abbf7dcd : SUCCESS in 6m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2285e565f1fa48db806072490576f27f : SUCCESS in 8m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee2fa7aced0d46ab8e6a58635c82f219 : FAILURE in 36m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b9909c524584adf879c48ac7d25d76b : SUCCESS in 5m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/945a07ab3a0f47d7a23e5fa454ada31f : SUCCESS in 1h 48m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e42534b70e264199969b67e27bec95da : FAILURE in 2h 50m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3af16493fa9d4d3caba3b85d124d0c38 : SUCCESS in 1h 55m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6b337f48bff04515ba8aac06d1f736b7 : SUCCESS in 1h 51m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2bf74a9aec86489aaf1da80bd7bd707f : SUCCESS in 17m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1edd3e262ef24e64874ed94c05d5fa44 : SUCCESS in 1h 12m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/35f9382b97ad4537ba254c1088bd4214 : SUCCESS in 2h 14m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e73649a98c3d41a7852095a404229d8c : SUCCESS in 23m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2dadb2deb1344513b62108820f6a00c1 : SUCCESS in 1h 50m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4ca8324a76434394a7a3529c324ed866 : FAILURE in 1h 21m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c13f5f3d106848cd930d02490dbe0050 : SUCCESS in 1h 47m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3e5598ea6ac4df38233a9356399b9f4 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"310aa02895a6aa90a6562f38e91aa5b36fbc13dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-11-26 04:03:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"430e74248c5202909564ab14f6b69910d8c3bd7a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-26 04:03:56.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/71/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f2f88c6036a0932b52bba1b6e29501cf12051798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2025-11-26 04:05:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6f78dc7b96ecab74669890bcc61f1b1ab0191c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-26 04:05:41.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/72/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ee5b4e1787e533788436eb008bcaa6cac64272d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 04:42:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc70350d24e4d7289b777898d971c8b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be95f72946a142299c90d2ce165f88eb : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4db233e53ed7443e8f39c00529cea669 : SUCCESS in 28m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4935c55e1964138ba5b29cf91ba4ae9 : SUCCESS in 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"caf3eb8a5faf7546e8010ef3fbd362dd7f021f9c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-26 04:55:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2577/.","accounts_in_message":[],"_revision_number":2},{"id":"fe11bc70e16f0740de31299d789ba60d58e3fa00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-26 04:58:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2725/.","accounts_in_message":[],"_revision_number":2},{"id":"c728f54a26ddc7270c01575f74d4c7fa5c339c6f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-26 04:58:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0175d5fae9284d219c023774d4e2de54 : SUCCESS in 50m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e1ce272d8ecd4a75bc00e2cafe837eb4 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":3},{"id":"1783c56ac3d17222d6c724cc96c944c90fa6c63b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-26 05:01:11.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":3},{"id":"9e5d04a1ed6e7e7690c95612a483498b43a67ef3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-26 05:02:08.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":2},{"id":"824159b6bf8acf007de348a37aeaffaecccd96b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-26 05:14:57.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968380 : FAILURE in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"47e68c81dbdcc0fed7b7a2f0660e5ab0fc1f9c63","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-26 05:51:40.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2578/.","accounts_in_message":[],"_revision_number":3},{"id":"6ecd9d63a2822cb2d312daaf1795ddf1c5ec66bb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-26 05:53:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2726/.","accounts_in_message":[],"_revision_number":3},{"id":"d00c6f10670e20ef9b168694e87e8e39e2ea6da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 06:31:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d8182ce97214732b3d07c02aa20c5ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525f20de8a954f2e9d050440cf8c2090 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bb58fdc465f4771b8ea196122bd9623 : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/852f05f21b14496da57ba0cb003313d2 : SUCCESS in 14m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f5e6ad40094f689b217109efe6ff81 : SUCCESS in 15m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/280e6f32de674ed999334c3379637363 : SUCCESS in 56m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b668613525e44ce1b838407a3acedd5f : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78f995eba86a4e23a851afa71094c05a : SUCCESS in 1h 37m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a21619d8b0f2407f8101305b7a9a7d1f : SUCCESS in 1h 01m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c61536cc84534c768265226a113c8bb8 : SUCCESS in 16m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d400da70d054300bfc4200dfb12feaa : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/90dedb6a97e7402da93e4ce9c13edc32 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3ca9e1a28bde49d38466beb48be23c56 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b9d07fd89004b5ab201bd2ee31ba624 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cd4800655f244cc2bbd1b21d9f7055b6 : SUCCESS in 8m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea05079978c44e8eba307cccfa7ab7aa : FAILURE in 39m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7bfabd34dfd94b78adb45cfa18dd00aa : SUCCESS in 7m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/72655035b808469780a64848d4d50c3c : SUCCESS in 1h 24m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/74fd0b0f8bf1426580dac6cffde59ea2 : FAILURE in 2h 14m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1f2824ffe84434cb1b3a5b018356f06 : SUCCESS in 1h 14m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a06d5a59ddfb413a9897e310210aefc0 : SUCCESS in 2h 22m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/92810b9793aa4979b3054a99348e1f7e : SUCCESS in 35m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be1fda4aaf454ec496cc9b782f2c2684 : SUCCESS in 41m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a29406e7fd8747528fb0e8444b869188 : SUCCESS in 1h 41m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5bbda408cf59405fb18997e84541c367 : SUCCESS in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5baad7f4b6d246259a78064f8414cd84 : SUCCESS in 2h 02m 05s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/16b777b8f115417dad49ac3b56a8c9c9 : SUCCESS in 2h 20m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f0133cab114d4d0ea06cf3bed3889e0b : SUCCESS in 1h 47m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd8e88c949a84f5c81c4c483598a63b0 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c35edecb3e076ff0fdf1a99989b2045f61e28f52","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-12-02 08:25:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f644981daea77b14ec0b34a8f61a4b207ac511d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-01-06 13:09:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"175246e8eeabed285d3ca0c1555303fc17749d39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-06 13:10:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6f2f242638cc4c54881f3e6a64b94e2a : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eeb8b2d5bc3649238bd16ea08d03655e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"e3ba59a7f8b08e8acae6ac57ff890e8a8109fe78","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-06 13:16:03.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_103/index.html : FAILURE in 6m","accounts_in_message":[],"_revision_number":4},{"id":"1b334cf90a9bdc7d56e6673592cdfb7a36ad5334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 13:32:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c5a0d30c9c84090ade62c26cb46733a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/776f3232b5144fcbb3d9bee954e3245f : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/efd9be9c3a044133993140de6413f321 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7db92574ccd34461b441b36aa8843cc5 : SUCCESS in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4a64d9bc1c387433958d7316fc3e12576aa91c12","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-06 14:44:56.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":4},{"id":"527eb059a5c414d53f9ccde9d14e8e4d86787c84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-06 15:20:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/29cdf882379d4c0892c0d9d2185f325b : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":4},{"id":"68e5db553c16b9cb525df95077e9bc6c616c8cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 15:23:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7649217fa2a5460b90ab37cb43becefc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a11e50e256524d2ca8961d463d5ede3a : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da159de872e74ddaba0330ea472e69bf : SUCCESS in 10m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/207c149c368d4e79b58bff61782f1b18 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f5112e4e1c4b77baa0838016421b23 : SUCCESS in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2724b572cdf9440583bfbeb9bb313c2e : SUCCESS in 56m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b231ed503d9429abfc83822039715c2 : SUCCESS in 41m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03d1923f417a42cda31019d33ac8acbd : SUCCESS in 1h 17m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e165a3da62df41af9bf7e73a9c18df17 : SUCCESS in 43m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1dc9dcbd82d447eb08ccf2fe6935ebd : SUCCESS in 11m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/719e24bf419949828c83155d1bf2898b : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8535fb6e5f6b4d8893841127d1f4339f : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ba2a9d4089874775afb5bd7f985d55c7 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eae0e834cdf145bbb57212b29de87f04 : SUCCESS in 6m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fad4668d36214ccbaa3c0daf3c275e40 : SUCCESS in 7m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c63409de5b3e4124b6acf6020c0404db : FAILURE in 25m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/639645ee21b54dd198f651479b72d4d5 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ecada23ede14e4898229c8a6f6b7be0 : SUCCESS in 1h 03m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e6ebe8b90d934ea1bd11d81edc1fd1f7 : FAILURE in 1h 30m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3eadd6b0b40643b59f95e7bdf849f3de : SUCCESS in 1h 47m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/de68ac61da254fd4b5d37e0cde482eb2 : SUCCESS in 1h 44m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4b75590ffaf4445a013774a404c0cd8 : SUCCESS in 22m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c7b7f89a9814318af1b351fdf5de2f6 : SUCCESS in 1h 09m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/250bb0aa48e54d85ab0bd8abbcb45605 : SUCCESS in 1h 36m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/da48d5d3f6b047888e0ad8489d292ca9 : SUCCESS in 41m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a1bb40c0b27c42faa838eb07d196f816 : SUCCESS in 2h 07m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/51a9cd7467ab4901a0b4f0a3e19da139 : SUCCESS in 2h 06m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a92cbd29524d488fa40da6ee2ca86a3b : SUCCESS in 45m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e967348891b4d5db568dfacf247c415 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9938646228028027996f8810b9f9117c778197b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-06 15:30:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_01_2026_15_23_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"bbe2d1f96c932a744eaa25ec8069ca69453a46e8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-07 03:29:35.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-4-fc : FAILURE in 1h 3m 38s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-4-iscsi : SUCCESS in 1h 8m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c188bdc0fd736b8a50d26e65f604cc875bb716f7","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-01-07 05:23:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29f7511f55d10ae08f12963516b7b0a8dd978d9b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-07 06:00:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968380 : FAILURE in 1h 20m 51s","accounts_in_message":[],"_revision_number":4},{"id":"38d0405099ce650ecd55424a6918411fc221cc1e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-05 15:38:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c4aa18114ca801e0b9d37cfe615ddb08f73c5d93","author":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"date":"2026-02-11 04:52:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"78112be2958ef0bf431e18a0f392711590a44258","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-11 07:55:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/fcf4eac/ : RETRY_LIMIT in 8m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d13205d/ : FAILURE in 1h 37m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"82143c75eb605546419dbbd09b0b60c915a3ca86","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-03-02 06:07:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6677f7ceeb4b8b30418356510d02e9c78619cd42","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-02 07:20:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e8078f62d814175e4a1cdcbd36c5e08feae9ae70","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-03 16:51:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6309d8310ebd9b07051de0031587fe441cf494b6","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-05 09:20:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6211a2e39389c30489a2089ec4f4e6fe571188a5","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-03-09 06:31:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7ff41f52ae1a702c9cc192562681684cc111a06","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-09 09:44:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/ee654ea/ : FAILURE in 2h 14m 13s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/41fec87/ : FAILURE in 3h 10m 30s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a7c9687d0d9dbdaf2d6e71453170ab0673b94dca","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-03-22 15:11:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fedc597b52c4f029e193b8c23038c7138e811ddd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-22 15:43:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-fcp/d87c260/ : RETRY_LIMIT in 8m 52s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/1342fcd/ : SUCCESS in 29m 32s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f58da46/ : SUCCESS in 29m 20s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a692defaff230cf3d58538ef94718598966da903","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2026-04-08 18:47:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ac67d6ff0700b1d4639acc56f43b5f7efdd4990a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-13 14:57:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"286c493768c32e1702cf5b7567b5e1b5702d5ae8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-13 15:03:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/342c57d241a845a79e95f8c5a5935179 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/079722045ff940d8b9e1c74c347362a9 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":5},{"id":"7d41a6d7e62f98cb9d91c8ef8050e158d69e53be","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-13 15:36:32.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":5},{"id":"6ccf911e3fa07cb7fa07546ba211d9122beeb265","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-13 15:46:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/76be3a0802234d088afd3ead6d0ecddb : SUCCESS in 37m 09s","accounts_in_message":[],"_revision_number":5},{"id":"dea254c80814e4a6f7073045790eac62a1e5844c","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-13 16:06:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/1be6dffe772542bab4b5e8b2f6b3507e : FAILURE in 15m 35s","accounts_in_message":[],"_revision_number":5},{"id":"e2c1f089139272508867d363451f7e3f1af3c2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 18:15:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6d25e4c5fd043a9acab16a1ceacd67d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bf8927fdfd410c8408e2d80ac745d3 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57ca90954cc94250af184ab4be54c26a : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95c5b9a1573f4aa990ba437c2a568b0a : SUCCESS in 10m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/230553dbab3d49b9b027a0a44e6519c9 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0facc3465f4479bddded9ab5776100 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/94ad48209e6b4ce8903d15be33d72845 : SUCCESS in 1h 05m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41407f2244984faca3b86e369fd47569 : SUCCESS in 40m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab83e4ae143142e0b0690a0799ced0d2 : FAILURE in 1h 19m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d4e64c816634758b15cb6ece4d29ddb : SUCCESS in 40m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9386302854394c928f7dfcf0c5bb7f4e : SUCCESS in 11m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba9ef2173d1e4c6bb02184c5445c3621 : TIMED_OUT in 41m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f72990e1d7ae4ae8a407930ae0d4813f : SUCCESS in 3m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2375f832d22447fa96d10da158024206 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea9832c2bc7841cda1653242552afbf0 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7efbfaa7acdc445f90bb9c146f6eada3 : SUCCESS in 9m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/50fce84a5bce45d5bd41ceccc870f4d9 : FAILURE in 40m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d72dc00ee1b46d6bcadb0c32afe35c4 : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a541d0e43eb49f0888a863e6c00b3b2 : SUCCESS in 1h 44m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0187b7b7a6c94a2e8513a71b9697f838 : FAILURE in 2h 41m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ac2b4c8d0f546d1b4d4a40fb2a453ba : TIMED_OUT in 3h 08m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/76a5000089fc40888a506683d88298b3 : SUCCESS in 1h 40m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f05f2841428443a3912daf5faab0bfd8 : SUCCESS in 36m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b62368de860f4949b85df97e9b09739d : SUCCESS in 46m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/908151c0c0344951b8f8c969664ac89d : SUCCESS in 1h 32m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe961b3179714ed9a9e4dd1595bbcfe3 : SUCCESS in 49m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72cc98315c6f424c9b0bf3e50cbde4c8 : FAILURE in 1h 27m 06s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/effd9ca8bbd447d793db75cd55a0ded5 : SUCCESS in 2h 40m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c0bdabfb3d94dfe8e6d6499a390c6d9 : SUCCESS in 1h 39m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/817c6bcd15874d80b9f8b96f5f1f0f40 : SUCCESS in 45m 41s","accounts_in_message":[],"_revision_number":5},{"id":"ca5b36daaa207eba0c867b43ab5e10d783c11f84","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-13 21:02:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968380 : FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":5},{"id":"28f5da8fbfb3436b518f89f5dbd918d5c283805b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 21:31:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a53cf10a134c429fb41cd47ea63ceaf1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a5ca00b9f4f46b9a9f9fd47ff1ab203 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c65147a89066438cacd79323054d73ef : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d97bfffc979c4d66b1218776e53aca40 : SUCCESS in 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b25312980410437dce18fefe6daf56b095945716","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-13 22:28:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/898428f1a40243b2bbf166c3635f7776 : SUCCESS in 46m 11s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"d9fa47199dcb7288984fe89f5cfcb80b4df42250","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-14 04:02:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2fffa0d951548adfb60b720240321ded2b9b5917","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 04:02:51.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3606/.","accounts_in_message":[],"_revision_number":5},{"id":"05f0a5c3b513b27367cb6cd87e279a140057ddad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-14 04:08:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e5055d4320e547f69de9c7e539a49e6e : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/82425862b79b40f29e7a7c38ad484684 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":5},{"id":"189cce4d5dc55ebe42913f68240e1ec5679919ee","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-04-14 05:48:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"40dce66edb470fa6c082b81debc167671a36db19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 05:51:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed3325880fff48a9a5045568e089c6ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc54fd4e6b94dcb99c5d5398c346e72 : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd17781dc77846c88d13151567b71e36 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1c7494ade034f2ab86592c31cb04d50 : SUCCESS in 8m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1805df6c6914b879b7d81fae5ee73f7 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27154c5f36b4f5ca52c7cc9ffae6284 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3add6c492e77459a96c0e7ff0889e872 : SUCCESS in 1h 03m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90ed429bf2c24264ba5587e5a43cddca : SUCCESS in 35m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a93b03d407c4a5d8d6a3563173b4835 : SUCCESS in 1h 35m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da2fd5f4f3254ee5a0e4b4b05a5bdcca : SUCCESS in 36m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4f47f549b914242b37a1a1a91064bb3 : SUCCESS in 14m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e9b9a66a869a4c1da966b95cfff88304 : SUCCESS in 22m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/280b44075f3142b59509c6509139318c : SUCCESS in 3m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6e909ed7a6c4ddd99c1e392d76a05af : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f2c7c89544b4114b7a137ee2419f14a : SUCCESS in 6m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2aa67455d49f4ea7a3139d47bf105490 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1e9a998904fa42d8a216071a12355219 : FAILURE in 41m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0dba817050c94034be40f86dcdbac868 : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d4447c568eca4bcaa902ec750f35db51 : SUCCESS in 1h 05m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a10249fd9aa4774a540cd1aeb7eec64 : FAILURE in 1h 40m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f663e7c2fafe443db13ba88a160d755d : SUCCESS in 1h 44m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8862597cb6044695af1f0bd15d1ec113 : RETRY_LIMIT in 2m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fa0f38a0a0e944e4986b0feb7cda76a7 : SUCCESS in 37m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f0d43b8042464ceeb1ae21f07dfa0801 : SUCCESS in 41m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d235b2f5adc449e186ececcfff734691 : SUCCESS in 1h 32m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4e73d995dc0744d9948d1e63488f57e3 : SUCCESS in 25m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8fce4d07d4a8403d8c6035c161107ba8 : FAILURE in 1h 35m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/cec23c85b9d348c6bfd69784f7c920db : RETRY_LIMIT in 3m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bdc65c08c40d475cb38843fba96f3528 : SUCCESS in 1h 47m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30e1b978f3ba4f7baa7118aae085ffb3 : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":5},{"id":"7b4e75fcdf2f5310e35d9b5dda21f3bd9bb944eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 06:51:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed490a515004538b6ca43987cc93998\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5789195323a4be298b2c38e32479027 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2051492880a147e48feab04c5172f8a0 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3629ac3d0f034ebabfd369b2c198a29a : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eaddca0a199a505a051c77632385861341625424","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 08:37:26.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3782/.","accounts_in_message":[],"_revision_number":5},{"id":"75adf6879a23aa22be13d922c77da78215c88b9b","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-15 15:45:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e857da8cca7654dc8fc77a40f8fa05e641e03c90","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 15:49:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9fb832a0c69d4a9db56e56574fe05302 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f269ca33ee7b4929892b90478b1fa599 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":5},{"id":"9494b779a99fb5c6c9803b8d0adb2bd05a58943c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 16:30:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d349ee7a7f64ea9ad72aba1fe7936e8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/332fbfa0061f4959ac0c03e17f20680f : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ac348d1bea54177a0688781a691d44e : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e5012620e3a1402e8640f2d5e8d1f2d3 : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a415f806fefa465566373d17c98129a10a353927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 00:28:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef65241c4894336a5a1b29b6619d73a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee1b62b71d64381b6f0a167a929fe50 : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3f6e5f5490848edb2c9588ef80a8962 : SUCCESS in 13m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba00d20ef6184ec6b282f02d1bf81350 : SUCCESS in 10m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd7f1d6c730845a38934cdbd064c9772 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b48d4ab71d44e680b18e5e0ee28a28 : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b70c58eb3b9945499803dcf40913c599 : SUCCESS in 52m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68ab147a6b0044d5906b5ac8f45c12f7 : SUCCESS in 32m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4c38dbe887f40e1b99419cbbd73fa52 : SUCCESS in 1h 28m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdd3f7b15de14401979a85b3ea4f888a : SUCCESS in 1h 03m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3900a13aa9846f8a5ed0116f6f51326 : SUCCESS in 12m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18c4e8ac5c9f4070a7541f94cf5178e1 : SUCCESS in 19m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dfdde8699c594c27853a16d34e2075ea : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bddb70acc03246ea9aab69fd69478ab1 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69dcff56a04f4dcdaf40cf34fdde35d2 : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fbc7c525b32248d0a74b5f3e21e00fbf : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/416da34174594e48a1719d9f99281693 : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1224c805c4d044c8892fb401999238d5 : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/784a13281bfe417c85d17191c41dc3cc : SUCCESS in 58m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e4d02915deb4a198a5334d8515b3286 : FAILURE in 1h 44m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d64dca466be940df9088817a1cb9128a : SUCCESS in 1h 27m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/eb43ea2e8ef7473e9c1a006300285661 : SUCCESS in 2h 15m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1014323fe6f94653bd7cfba5444e6dca : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b36799e50fe2412995518f40d31621ef : SUCCESS in 1h 18m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4a2c28e9d7e54fc0b9646a0a2b8c9efa : SUCCESS in 1h 22m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/55bca0320b79483a8e6d90574c4ba19c : SUCCESS in 24m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1795a3e12cb4851a1964fe4b1f1dfe8 : SUCCESS in 2h 25m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a70ce3f0955442dba45bac27bfbea78 : SUCCESS in 47m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a80e4176a99948ff94f0c30973857db8 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":5},{"id":"c9f4240dfa7c59ca260dcb17c5c5c335324fd686","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-16 00:37:24.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_04_2026_00_31_19\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"e9ea02f14490eb2a108543cf1aa0526e3d4fb03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-16 13:41:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1848809c211ae408ea7b46df5f71c199e11bbc80","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-16 13:41:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/063667d6767745a68b16de413fa66606 : NODE_FAILURE Node request 200-0000707328 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"9542f9b85c3bb8bd223a0f0a3fa2e45c82973139","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 13:46:37.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/575ed20d99b748908de44b9135ecc946 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3e47df5ac9474484ab69a0a7ee8b7fef : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":6},{"id":"4e0e2e938a29990fd5be56ab2a59d7fce9f6e60c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 14:39:31.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":6},{"id":"436fead9339a094b91db0e2a50e50fdb50e6d125","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 14:47:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3658/.","accounts_in_message":[],"_revision_number":6},{"id":"fe4d41bdc5456d3f3eb84065631983efb7359973","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-16 14:49:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver build/711be20f819c479b98d0ca7f5f492fc7 : SUCCESS in 1h 07m 33s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/968380/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"544b91634375f5fcc37b4c736a4e32bf281f1cc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 15:08:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63075600a8bb4e649f9c381ffa483032\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/752b5107467547a58bd454af278c6fdc : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b7c3314a13b94a4c9e98e983d6f76466 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4ab496fd93f943fc98dba0709404076c : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"841c675c2fd5d623694e379fb05877e139e3303e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-16 16:23:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f5740fc1d638448bb101ed7aa6dff93b : FAILURE in 1h 02m 27s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"848ff55d9799e3c8f22ad94b7bceb42d55ec8d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:53:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f62eb0295da4881aa709782980e7405\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef903f1a235440c992b4425d0dfc3add : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e4a7de5d7c341c48d962643ea4721ca : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/618d1206966c44509661382f80ce161a : SUCCESS in 8m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc97f271d5764efd9bc639dc55e11bd5 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcb0b412dc047318d1f8765be90cf92 : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2ed30562e2ad4139a33b8942ae883689 : SUCCESS in 31m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bff777b0e01147fabd95dfbf951fd87a : SUCCESS in 1h 05m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6943614f81c48fa8e68a2ca62f38bd4 : SUCCESS in 1h 58m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25b62d6ba78041ee85ba038704d03952 : SUCCESS in 1h 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e841ec86fa14219b358d224d145e47a : SUCCESS in 20m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8f2a23b8a5148418facbd20d8daae28 : SUCCESS in 17m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b78b52d6f22f4b33b000db766524eb4e : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/71ec04dc14f54e618d4d402b3082fffc : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b397e0ec1174c958dea0ea8cd12fa71 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/20c66ad66ab742018bdacf68aed07901 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/556dd9324fe2461bb87f123fa8b7a784 : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e96dc1d22fd04fb28e38726235ebd16e : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3f3cb5e680d4c88b484952878ae8197 : SUCCESS in 1h 02m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a6aa0fb29a14e758af6e857a92ea531 : FAILURE in 2h 29m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/152aa044303d4933bca68005c79f13ce : SUCCESS in 2h 10m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f6c6685a269b4869bb3c835fc73d9ea7 : SUCCESS in 1h 54m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5309f1c464244048af3d233f8167543f : SUCCESS in 22m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9afccdf7e9174461bc29f62368c68652 : SUCCESS in 1h 16m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08459ad5f11043cf93f008a1b0e6eb52 : SUCCESS in 1h 56m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f0ffd1458ec64b6681dd4073d1a37fb2 : SUCCESS in 34m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db35871d67994571bf973d67ae870834 : TIMED_OUT in 3h 07m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa2de8d74b7b41ac8f643e25f4787f37 : SUCCESS in 1h 54m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9e5e2e1761c497e84ebce6b2455389b : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":6},{"id":"b02a6a6cf094b5dc4dbafc28a9e38f1428330989","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-16 18:20:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968380 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":6},{"id":"e733f12e2d061b61ed20ed63e143fdeae13af205","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 03:45:58.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3834/.","accounts_in_message":[],"_revision_number":6},{"id":"70681f11452de7a20e18355be98393d8600d9a87","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-17 04:15:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c494b8b1b402904d19668a5b77bb6ecb9cfc8174","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-17 04:21:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5f14dd69e07549f998c72a3aecf1586b : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ef415f3ef5c346aa8c225e9dba204ab4 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":6},{"id":"e331ce64d95b88f8e6cc3714d9151ce838fe14dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 05:23:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbec3ca4ff624c1cb75eb2b2d4698c79\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66ad2887d4b94f3ca97a55d539c5685f : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34a7671192394e6fa561b7f533a2fb5f : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/652a4cdaded947e1a3ce05ceb3e6c04a : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96041240b8159f46c2aca79910af60edb3f18f5f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-04-17 09:45:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52671ef2931e88903a6d160eb13039f15969da92","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-17 09:50:38.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/639389bb37384c278638012c797e3bb6 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8d45c94504c948e99db9ec22bcdf1f04 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":6},{"id":"09bd85d7fa343659e8e63c60261b7ff76b5b46e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 10:22:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f213760079624109be5caa3e865bcee4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36cdad1cd3704be091583e060a4ff0a8 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0f2839441884e4fbc1c3d9eef3b635e : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/61508ce3d09e4c11ab23300f819164f3 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"18d7256e5923fcdfe821106bfca48cb0db9f051c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 13:41:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60e02ba7058348bc8a6caf7e6fc7d0fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a86398a2c1944beac70700040531a3e : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e9ef0d323604ba4a05777bcdb275a11 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8748678cedf8476582d410dbfd36b993 : SUCCESS in 10m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d8aaea1510f4000a976ce9a1be7fb4e : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ad69281507430ea3fad9a8c227e6ca : SUCCESS in 15m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/bbe585d06dff4f6fbec96a104f1af2bf : SUCCESS in 1h 09m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17890614807a4e6eb9e22cdb73c7d799 : SUCCESS in 1h 08m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c62cc03aaa814e2c929df1663f6cd416 : SUCCESS in 1h 30m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dba7526a564b467eacd92204fbd0a048 : SUCCESS in 1h 09m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e142f614cb64677902348d0635d3d12 : SUCCESS in 13m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff6a6cb188434122bea8de4ecc5c2d52 : FAILURE in 24m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3538ab673af9412886ad3ebc5ed4a23b : SUCCESS in 6m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80b2d3fad79b433a85610f70d70b74d1 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b0d5d47edb94048ae1e2a3367072ff0 : SUCCESS in 8m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/36bf1217278e4e2fab8550a58202d7c4 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d0bf5cdf0f9043428d8f2d74064437ae : SUCCESS in 41m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39f3bcac2409435396a10e52781a7df7 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/89448829c5144115ab322599215bc3dd : SUCCESS in 1h 46m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/749af5f3f6a947458cfa5d78006dfb85 : FAILURE in 1h 41m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c599b0f566e40a79cb0cfa8003021ae : SUCCESS in 1h 43m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6e40add5db9744659ede654aeb5f4a07 : SUCCESS in 1h 53m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a30506a6cfa409b80079e1d0e6e293e : SUCCESS in 29m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad54444992734d688ece98a1d6fabfad : SUCCESS in 1h 16m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a94782d8412745be8f83023c55208238 : SUCCESS in 1h 56m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b66779b4eb6943be82dddd1de59c8009 : SUCCESS in 47m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4f638a5af15b4b5cbce576757d223af0 : SUCCESS in 1h 33m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85f492b7f51c4bd6b6c326c2423c101d : SUCCESS in 1h 37m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d343ab1f874c44f7924a0f0c6d349fb5 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":6},{"id":"b93d787a19c863127b6bf1a0b4b6bc7f22624164","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-17 13:51:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_04_2026_13_44_37\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"185d3d359a5846caf82846af34f27bcac089ade7","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-04-17 15:26:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e0746b53e5693b818aa58ed4bbdd1726dc446849","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-04-17 15:40:13.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"34cf5653cf152ad7521f71bd7b4335ee0553ca78","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-04-17 15:43:06.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e2bd6bb4cee11af186a71664d433d65d7615a7d8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-17 15:59:08.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/6/upstream-check/cinder-tempest-plugin-ontap-fcp/7603559/ : RETRY_LIMIT in 9m 12s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/fb289fb/ : SUCCESS in 30m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a6655f7/ : SUCCESS in 30m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"2b9bd8351fdb15a68baa2bc10d4f9ebc61ee6af3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-04-29 17:46:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3bc81ebef73e7c569b200a1bc64dfd69ff065369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-09 05:02:04.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"46493b4f059878cdca6d0792c5ea29537c233ab6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-09 05:02:15.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/882/","accounts_in_message":[],"_revision_number":7},{"id":"d973743786216ba1b02011cf820b4f23ffc996b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-09 05:02:19.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/968380/7/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/968380/7/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"2d1891e81fc09cd19bd0f8752cf32326ce23cbd4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-09 05:25:53.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/63bb1b9a9e824a97bb81e5859398ea85\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf6fa1df3c6043ad891e8506959d6bf8 : NODE_FAILURE Node request 200-0000064267 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95f930bc06ac48d39b441d50937220be : NODE_FAILURE Node request 200-0000064268 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ceef0d7f13d04178ba39aef5efd33e5c : FAILURE in 21m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2506625ffa444460a2c19ac5dfecb310 : FAILURE in 23m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/32334bba2b6f408183e4591f71d74ff3 : FAILURE in 20m 53s","accounts_in_message":[],"_revision_number":7},{"id":"34ba3079382f0a553b819cddb883718cf6357010","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-09 05:40:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver build/b24429b6d89c4e0badaa60c2d76f8a39 : SUCCESS in 38m 21s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/968380","accounts_in_message":[],"_revision_number":7},{"id":"1f081f01711973a1d4d371ae171c582a6fbe6823","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-09 05:42:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/18b84eec21ec44f4b28464914471fe0b : SUCCESS in 40m 09s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/968380","accounts_in_message":[],"_revision_number":7},{"id":"f94fa0a7ec7372f7d9ded49ab44a5bd1200b0aca","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-09 05:49:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/f0a4f56268a54da6833a4c005545c4b9 : SUCCESS in 47m 12s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/968380","accounts_in_message":[],"_revision_number":7},{"id":"97f54e0d812ef5d5938bd5c1176468cea3cd308b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-09 06:00:57.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":7},{"id":"7b20b097fd179f1a9d7f3a0ebeef0465aa547603","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-09 06:27:50.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6eda2112ee79305145041776219c931a981de9ac","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-09 08:07:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/724e44bdbfce4b3b9332146e3ebbfd5e : SUCCESS in 54m 17s\n- cinder-storpool-tempest-iscsi-multipath build/a5d670e13b3a43469378c2f5259dbc00 : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":7},{"id":"e8ffeeb6ef178070b0bfc299dc38bd6631bb130d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 08:53:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a46985f4a4446dbb93c6f0f9b2ba915\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490513de88664dee810da31eef3eefe4 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9e81939f13d4c5583e56715bfd2ab87 : SUCCESS in 10m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/663d279d33504b22b8fdcc01f741375a : SUCCESS in 8m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbc03acd09d24c8f95ac84277c9be217 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e7c435578b747bc95e1d853e3e682b1 : SUCCESS in 13m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/778a5c1a5dae474599a9c449e8a8712f : SUCCESS in 1h 06m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fedaaad62dab4dde81769fb8f753930a : SUCCESS in 1h 09m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/000e7093b3b04d4ba59fc72199d6d8b1 : SUCCESS in 1h 44m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf29d80af4df46f38d9a1adc6c44d08d : SUCCESS in 37m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ec3f40b263442db9eadd37fa1228d7 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d8f483806a2949b39220d0ccda3098fc : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/267e1149910b4472bc0f0122111f9cb0 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a7532789faa4d41a9c6bfa3b75cd1ca : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ed7d844240d4fe78d15507d722c0293 : SUCCESS in 5m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/547052606493465585816b7705e21880 : SUCCESS in 6m 31s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f58344db44ad4578bf09628e334561b0 : SUCCESS in 12m 04s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/427be074599d4c07bb334ac2b04e300c : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb733dbb531146ae9c072c2d30f3fa7f : SUCCESS in 3m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/45e4a69b486e4329ae8e1cf727a4055e : SUCCESS in 1h 03m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/858f3be94e0e405aa8a4f13b3cc786f0 : TIMED_OUT Host unreachable in 3h 45m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d915a6d657f4fb8998d9509de782841 : SUCCESS in 1h 52m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/361204f24d454232a6033be9efb19955 : SUCCESS in 2h 02m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fd25d1d2ae2647db9db4fc0938376d9f : SUCCESS in 31m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a9340a5b11cf4ba591e0ea3216423a08 : SUCCESS in 38m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d41a967a40da40d7b448b096055e2f70 : SUCCESS in 1h 52m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0108073f75e24e2bbba68b47e4a5fc6d : SUCCESS in 45m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e35604fcea6c45438adb2bd763061364 : FAILURE in 1h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c274ae646184fe68e3ff40154ef9781 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":7},{"id":"1d1f930a6932c61ca5277a175cc5adbfba57ec4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 10:13:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d6d05c5f5c42b3b9a07d91aacc200b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dfebc1cb6aef41c38026cec66aa4f667 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9660e83ef19a4dd0b0cf9103579a2a11 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f18124a440f4f378e0d5172e312f11b : SUCCESS in 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd0e8bf72b70c50dc61e3ff22f98daa50d2f5afd","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-06-09 11:01:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a6e9194f2e3dba9fe957f4966a0420fd0b88adea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-09 14:08:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4404/.","accounts_in_message":[],"_revision_number":7},{"id":"6043ba6e1645b2e5a1a7198b11c263f37bc7f787","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-09 14:32:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5e0352d96e1a69d0b712ff5aef962b5caae1a2ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 15:10:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cdc0e77e7a840aba4efae88da87eb6f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/516157b8a5aa4c208d58e76c9dea0bc4 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a146040e51b46b08a36fe0363a83a16 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0fdcc2ce0a764f50b3c5f32d4158a9a0 : SUCCESS in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"071953353420ef7163f1725ae255614ef77dd981","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-09 16:33:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2965be8c03a34cbb8a264dbb75ee8d86 : SUCCESS in 56m 11s\n- cinder-storpool-tempest-iscsi-multipath build/162757f778924307bfef608194d20daa : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":7},{"id":"ad390ce4f6035fec0ebf8a844766588a80ce1e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 18:23:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793ac633d7df480eabd0b7389b80a44d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c76ac477797473fa1241225d311a9cf : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ce1818286d94f718a24c708a68d1a01 : SUCCESS in 13m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dab053ab37e64f19a8ef2855831e0631 : SUCCESS in 11m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed75fb215e8b4da3bae389812db79454 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e2e7874cd64e2ca101139b725d5b4e : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/31761aa407be4f94928dd9c893e54b7a : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/920624f3376846f4979db3ced1abf5c0 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ba15ee76fa64d5fb223e91bd6c7ee0c : SUCCESS in 1h 34m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdf387e2601e4554a48d5c48c596dff1 : SUCCESS in 1h 08m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1e8bba12d6f4a2893141b9f73df6158 : SUCCESS in 9m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b618aa02e51943d0bfb2a6611ae26554 : SUCCESS in 16m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ab17cae3b534ef1a5d59ee26872c1fc : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03f73dfe2fcc40bfbaeb3caf71bac5e4 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a08a7e13aa7547948b173d43c23d240b : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3eb020056a5d4705bc828cc8f79c51c8 : SUCCESS in 4m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/64b4127727f343ed9273f58119e25523 : SUCCESS in 8m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/46659c72cedc46d09f844f21c0d29a30 : SUCCESS in 42m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe1a5e31172641a291a8f07a730b0081 : SUCCESS in 5m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bc50db04d56a403e80ee80b73469542a : SUCCESS in 1h 54m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5c158ce31c6e4cc1ae657460cc7dc26d : TIMED_OUT Host unreachable in 3h 41m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7be42bc2b15f498ba3d4466ef9e8f8d7 : SUCCESS in 1h 52m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4245d494d3c144338971d9a06cdfc4db : SUCCESS in 2h 09m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2d4e75fd43fe4f4fa712fbfcfefa75b1 : SUCCESS in 36m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7a1f44d57b704b7c9d015ef09e065120 : SUCCESS in 1h 18m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ab746bcec8f40eeb384644f5306c4f9 : SUCCESS in 1h 49m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef4f4d4cd4ff4fddac00ab030211c24a : SUCCESS in 50m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4c3e39f114af404cb0a8e2c87e0ef79f : SUCCESS in 1h 50m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06abe7668afb4e66a1a8311cdf7b943f : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":7},{"id":"5eb0b083471ec8411d6d19b904e20a107958a7c2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-09 18:37:36.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_06_2026_18_30_20\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"7c7b6c2e89c968c6b9431982f5301da545cf779f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-06-10 05:07:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8d32c372f41c714922e115f481ff817001d4c95d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-10 05:38:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/7/upstream-check/cinder-tempest-plugin-ontap-fcp/460a758/ : FAILURE in 27m 24s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/c0a2128/ : SUCCESS in 27m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/968380/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/78b7f67/ : SUCCESS in 27m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"0910f5b3d5644883f2e4c58ada4b4505d88dc7e6","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-11 12:04:01.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dff7a8c3f70f72ea08a34a263cb8098fc847f88e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:25:25.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/883/","accounts_in_message":[],"_revision_number":7},{"id":"10f7873c45df892d3cfc9407eb1274365acb01d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:25:27.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"a988d1243d0c19228f32183a05dab59077243578","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:27:13.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/884/","accounts_in_message":[],"_revision_number":7},{"id":"25ebc5b631a279b9f96a4c565a36ed2b59bc18b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:27:15.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"e4ca53eb171f9cd6014d9c3f04ae5b72f2c6db68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:27:32.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/885/","accounts_in_message":[],"_revision_number":7},{"id":"932142a7a424b3b8bc34545b393fd429a496db02","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:27:36.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"3c27c4a42cd1a0788145216da9940bd1b73f75bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:46:21.000000000","message":"Patch Set 7:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/886/","accounts_in_message":[],"_revision_number":7},{"id":"01e8308d12a22971af68c261a1cac9f61a71485e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-18 06:46:25.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/${GERRIT_CHANGE_NUMBER}/${GERRIT_PATCHSET_NUMBER}/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":7},{"id":"b574c323bd11a440a8d1b2f42a457d32bff2c9af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:34.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6},{"id":"c333833262c544371a8b91ead580aef5d0987405","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:56.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5},{"id":"6c11898df36f76200ba54f14ba1ce12fe8ad8cfe","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-06-28 04:16:28.000000000","message":"Patch Set 7: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"90f6a484ba55d796f0140b8bec1d90008432049f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-01 16:10:25.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"01324af3ce2c83cc84566427e2aaa250c177b929","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-01 16:11:31.000000000","message":"Patch Set 7:\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8c2774d2562b7eb7f320836e9bf65a99a3ce44f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:37:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8801e6e54be4cefba98df8ad4d951a7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6911cacc3074e8f9758147305215b3f : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a63f8cad70ab46c3b75e1c3f05f74378 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/92f05cbb6e4a4f53a7bbac296a1476d7 : SUCCESS in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d1af25709e4bee3e2712aec7f18e147eca025776","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-01 17:00:38.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/66b6942bbf6b4b48bbaacccec19aa960/","accounts_in_message":[],"_revision_number":8},{"id":"5d23ab848bec63e60dfeb7c2f8ade81e4d4a289a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-01 17:06:57.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":8},{"id":"e361919268cdc73112be4dc9c89b7afc326cc519","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-01 17:34:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/28946c28a553424d899465415540a3c4 : RETRY_LIMIT in 16m 16s\n- cinder-storpool-tempest-iscsi-multipath build/f99380bff7d749cbaf5294b492061bab : RETRY_LIMIT in 16m 53s","accounts_in_message":[],"_revision_number":8},{"id":"3fd7fc85d793d00c1214e1b2776c186ce1893302","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 18:06:21.000000000","message":"Patch Set 8:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/341 : SUCCESS in 0h 42m 36s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/968380","accounts_in_message":[],"_revision_number":8},{"id":"323f4b08635b0b8925717afad85ae8900bf69cfc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-01 18:42:31.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/66b6942bbf6b4b48bbaacccec19aa960\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f024af1a4f714029b45227676824e128 : SUCCESS in 42m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e7dafc034d234610ae1eb60895eebfac : FAILURE in 41m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1dd4cdc2d074401ea6001382c7b825b4 : SUCCESS in 35m 19s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8777813cc9d4ee39ca47286aee92230 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":8},{"id":"e68f8acba8392a11953f6825b374cebf61100c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 19:30:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41eede965ae74ac2919d5373c7d08187\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c40de55796045f6a61ce0b0ccb31d65 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4c50f16cf2f4003bb1da19bbf5f39b9 : SUCCESS in 6m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59dad6b382a24c2495f4349fd57ff1e7 : SUCCESS in 6m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/47f6dbacdac34b14a0082d63f0377714 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8014cfceebc4efdbfd78235fbccff4c : SUCCESS in 14m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8034a31f4cdf4d8391805cbcea62762b : SUCCESS in 1h 06m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1c83edfe9cc47069fcb4c0682d48753 : SUCCESS in 1h 00m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d887b16c26d48bc9b98457349ea3be3 : TIMED_OUT in 2h 06m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87b7108000f44244bdc2ca3b830b1ceb : SUCCESS in 1h 04m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd54513d80d48b7a8ccb7b125415d58 : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0938531515e041278587fe3ee36529ab : TIMED_OUT in 41m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c67e2aa6f75477ab531e1f29f02090b : SUCCESS in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6c95165a84c4a8f809bcabad4703a18 : SUCCESS in 6m 16s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/30b9d34aa0fc4a7eae2c324e3157408e : FAILURE in 10m 51s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/84c7fec9e34a4d839d2227a0db13fb25 : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f41f5097cdef4b39b4675d31378a976f : SUCCESS in 6m 09s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/17d4c92dc8c6404ea4d8f0410fbf90ed : SUCCESS in 9m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c7d0dfd2f2294c1085d521399dde9e7b : SUCCESS in 27m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d0340c0b2d542b28bd233ae69103dd3 : SUCCESS in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d8ea9fe054a464ca1948c38b9a7be9c : SUCCESS in 2h 04m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/34db324fe409475cab146f971d110e59 : FAILURE in 2h 25m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ff89c1155e44325a622e8ed303f83a9 : FAILURE in 2h 03m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/16c8f1ee57ec4e2282295afe4f5df0e6 : FAILURE in 3h 13m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1a91a566041a4dbc8be73d3089e3774a : FAILURE in 1h 33m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1864e28f12f944658e6f18cb7a2d941e : SUCCESS in 26m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b8b588d8168443e99c2cd54e2396915a : SUCCESS in 1h 39m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/130989bcd66b4cedaa4f2f4ab89858de : SUCCESS in 1h 28m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a73ddd681444715907fdf255c0ff53b : SUCCESS in 45m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/215732d8735b49a098ae19004426992e : SUCCESS in 1h 46m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/237aa4edc1f74d08b34df4d0ac8f3a65 : SUCCESS in 1h 05m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"a66a081954991a381b316b7874c22cc19c287e17","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-01 19:48:29.000000000","message":"Patch Set 8:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/427 : SUCCESS in 0h 41m 13s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/968380","accounts_in_message":[],"_revision_number":8},{"id":"b4b187e4e8baa216b9ae500afc5fe731f13ccbc3","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-02 04:22:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"80bdcfde78776663f8341a72199fc150836fd22c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 04:46:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/099b7225d82448e683fb043cff054b17\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/954342cbba9742f488e412440505a348 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/529a9cc7db2a440abbbd4c5729f7dc03 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb607b4da3334647a618ada03bbc0fbd : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a026200d4a951d1cc9245a12f3aefde059a31cd6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-02 04:57:43.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b7490ad2953648fcb7d9e0a412ad5994/","accounts_in_message":[],"_revision_number":8},{"id":"002775f3be85aba836eebb221844121eee8d8c0e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-02 04:57:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/4d913c3dd7db407eb911943327ce0aee : RETRY_LIMIT in 16m 00s\n- cinder-storpool-tempest-iscsi-multipath build/6ed92e556afc4a00b43d551a7253a050 : RETRY_LIMIT in 16m 56s","accounts_in_message":[],"_revision_number":8},{"id":"d945bba94de706fd32053c988d4c6b47fe556569","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-02 05:04:18.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b7490ad2953648fcb7d9e0a412ad5994\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fdbdd4441e0a4ffca53d55f9a37cf108 : SUCCESS in 40m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d88f5f26e6304655926bc76210e9159b : SUCCESS in 35m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad1c9458edac4a7795d66fd2c8f724ae : SUCCESS in 34m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e52289b66b6847d9b3a5191ee2aa2c1b : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":8},{"id":"197cf85c0641d304397f407ceaba595c5e7ee7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 07:32:36.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/901d2375f63d4b7a85534d2458d0fa97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ffa483583bc45c78e5429bdd41348d9 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/051f2385dedb4e0eb7b1d184e9081469 : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45a470346f6f4f3fad4711a23d4917b5 : SUCCESS in 6m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88674bd40aeb4205a785179fc580335a : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f589770550436f94388ab9cd96deae : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ff6694c218a949e7b32626fff4bb072f : SUCCESS in 32m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90acf53cc1874d2fa779d91493ed023f : SUCCESS in 31m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92701d804769479fa13e88003046686d : SUCCESS in 1h 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d9651de1c14b459f55c99746e642bb : SUCCESS in 49m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e0178a023d543a0b5debbe615e9c849 : SUCCESS in 7m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e01749601794397b830b5a5a7751b41 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e86795b7c7e4ced99efdc49152aab9b : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f50efde96e64fec9446fc38940003d5 : SUCCESS in 7m 37s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/b5693c21f9ac4000a220317abe2afd1b : FAILURE in 10m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d544b6575054222a98af4986184f636 : SUCCESS in 6m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d0f32fceebaf44a2a4d85f3666b79304 : SUCCESS in 5m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6318980104ad4f17bccdb0584a8a9c70 : SUCCESS in 9m 42s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/410132cfc29043f8a508aff23f2c2cb6 : SUCCESS in 28m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1983f49971d3433aa065401ad87337dd : SUCCESS in 3m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13cb178a18b24f66b5eb57b302258a66 : SUCCESS in 1h 57m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4be685486df048eb98ae16a5bb025108 : FAILURE in 2h 27m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37270151cec541baafc5e16eb5e37203 : FAILURE in 1h 54m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2062eb80b3c948b0b99bf878643b70d8 : TIMED_OUT in 3h 08m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0d7870dd044f44669c21e07eaa4e5c73 : FAILURE in 1h 48m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1834a2ae5e024fdfb0b0c7189c4069df : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b0ac13632d374117bd49dfabb3b6a576 : SUCCESS in 55m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d4c0ee0db344ebc941fcfc694a343e8 : SUCCESS in 1h 36m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9911685d91e4d599dd48a6775932074 : SUCCESS in 45m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aeea7ede3f8e400396faadf505c1e719 : SUCCESS in 53m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25c69d5303a74cf6b025a5f333b3f0cc : SUCCESS in 57m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"7fb883eb2db844cdc867891d05a768c9783c010e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-02 08:21:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4618/.","accounts_in_message":[],"_revision_number":8},{"id":"bbf80d5096cb1a10bf180bee3174bf14424592ce","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-02 16:21:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fd6ca9aa556525d41b29f4e926df712a17056380","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-02 16:59:16.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/406e43893f1941a8bf22e103b0a78150/","accounts_in_message":[],"_revision_number":8},{"id":"1fb6403c38355ceec44584b003500c940cbfd821","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-02 17:21:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/70a37c2b43b44fa68d329301d7519123 : SUCCESS in 51m 50s\n- cinder-storpool-tempest-iscsi-multipath build/74cbac66fe844d1e8213147e40050574 : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":8},{"id":"593eb20aed5aacfbc15220409d1cf9ea395275ec","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-02 17:34:58.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/406e43893f1941a8bf22e103b0a78150\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac0b528238df4686b2f48866990931c4 : FAILURE in 43m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfd6b1513a41491d9b334b0c64d9039d : SUCCESS in 39m 13s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c05a411693e64f1183c8b70193bcb385 : SUCCESS in 37m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46cf67d17d64473e8d095c8430a11535 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":8},{"id":"ba71a65b7f724e2ab3357516838001089f55a154","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:05:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3039416d1760497c8dc629f9f4ebf35c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0386c40cee2643258c62d334e59ae517 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9fdf19b82524e918603cf04d982a312 : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ae4fb38e5a05467d8e6de25ef75b2349 : SUCCESS in 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"39113cbcf0d33d9fbd15b6f0a19811f8e917501f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:27:32.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4385c86940941d888544b5de0d3f511\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d67df486ba483ab8cad7caaca5433f : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d091b4ad62f445edaf6c0da4f701b086 : SUCCESS in 7m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e4de5a9ac4f48f9bc7aef0ba9b6c8f1 : SUCCESS in 7m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26f65477b76b4b799ce7725b78a8862c : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2f3f5a73dd41de90157fe5d5978a1a : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e907fbda586d419990b20c077f007b5e : SUCCESS in 55m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5000c920480c4609861c079c5329c16b : SUCCESS in 52m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c606b1147877495893d4575a17be04c1 : SUCCESS in 1h 24m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a91c64ba63a4ac7a4fb4294ece15bd1 : SUCCESS in 1h 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc609cda5bc7487d9f51628468642200 : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6a10b5e3246f42d0b25827e5e49f7cf2 : SUCCESS in 26m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1aeb9d3dd0b44b8f87e585fd1d1aa661 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/68dfc1e328494c6bb4ad2b93db30a41c : SUCCESS in 4m 08s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a11fecb16d884e20a058d10c21849af1 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3804433eb90b4bae96d70427e0de6073 : SUCCESS in 5m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/590f9f2a57064628a4bf7d0af26c42b0 : SUCCESS in 6m 28s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ccfc84991a6a4d53ac1805e2cf1fee76 : SUCCESS in 8m 48s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c050440dd66a44d886b9b0563f5fe2e1 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/350a27af45124d02b69ee2b6033fd970 : SUCCESS in 3m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/db904dbf5f7949749c1381abcf34698b : SUCCESS in 1h 55m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9896fec858304c72881325b1e1b23918 : FAILURE in 2h 44m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9a2bb759b0b4410585013c383ff0d68c : FAILURE in 1h 53m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e1743f10ea8c4bf1b7e794002f171b5d : FAILURE in 2h 52m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f071c3ce5c38463e9c710d69b5ebcea6 : FAILURE in 1h 56m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6d0c72949b9e47e5aff22b9ac1d5afd2 : SUCCESS in 38m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/77b0bbcafc8342beb85e0ccc80bc3706 : SUCCESS in 1h 21m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8557a0d1326547ecb4fbd766e8fd6c67 : SUCCESS in 1h 36m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c771cdb8eff847db83653e8ab24027f5 : SUCCESS in 26m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57837ccbcf524bcd99741b7c5f56e8a7 : FAILURE in 1h 41m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5139c6ddf7f24be28876a9638bc3b763 : SUCCESS in 1h 07m 51s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"268a8f7865e3435c42b8b41962e068bc76fca789","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-03 10:00:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a45b8ff2747a2c7cc97e3b14d6840cb8adb1bacb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 10:25:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae8d162b163f4c8eafd8e053d4ca96b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/637e814733e94975a039c03da117ce7a : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05726a54b6c14235a39ac0d6eab62305 : SUCCESS in 23m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/553ed890bd7743a290bb710c23864407 : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"71949044d744e03a42183b7be1b17786953d761b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-03 10:37:14.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/6603746a155244ee94d6ae8c43c1a444/","accounts_in_message":[],"_revision_number":8},{"id":"09ac033808f050f9155549433b0e38bd7737df63","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-03 10:44:57.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6603746a155244ee94d6ae8c43c1a444\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4a04ea030e1f421b8ac7371bb2e89d02 : SUCCESS in 43m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0f9fe7b0d2694731aefa713340e0d7ea : SUCCESS in 44m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2efe0223980f40fd87ddcd36d5fffb02 : SUCCESS in 36m 24s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69913601592a4957b5f2a53d375faaae : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":8},{"id":"d0eec0d956730750e0b1e1fe9db6ed2c4d6da07e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-03 11:24:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0de55e42bf894b6886570ef849fabd8f : SUCCESS in 53m 38s\n- cinder-storpool-tempest-iscsi-multipath build/b44b31c022a943b4b648838a0886eca4 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":8},{"id":"62aeed3efaa6aba9611acb55acfb8dfb66e87398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 13:20:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c500e4324543bbae219b5a2c006c0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13fb84e4a06d49248ed73a5c81d96638 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c432227d4fb4a0984276f568810aa91 : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8397f24c2e1e4f9c9deecc317ac245c7 : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6bf5de6cb12f4260904e1c7f462058b4 : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59be99a8970a488cacb8544dcde877e3 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/4f4bf2bedd554c159c76657d8fcc0f4f : SUCCESS in 1h 11m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6782bfda84ae4def830abcac7df02bc4 : SUCCESS in 32m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13d81e8b0659485b9e19b739d95a8560 : SUCCESS in 1h 34m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecb20e22cb454a41bc495772715b599b : SUCCESS in 1h 01m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ab5fd0bd37d45318e388a30df88ba1e : SUCCESS in 8m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fcd11a5a06247f18b94b899ba01630c : SUCCESS in 25m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d9d3e8ab86bf4141ac473259664ec6b2 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d60877fccd045458bf257e8c3c3cced : SUCCESS in 6m 16s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4220e55750754068b363eab33cb6a224 : FAILURE in 12m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1174028cb53b440ba150af0e407e3178 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/38620e46c679469f82c9f25938f8550d : SUCCESS in 6m 36s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/48ed5ed2b2514bb28486f65a81bce756 : SUCCESS in 8m 48s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/80347b42dcf24eec8e1859f324ea2c22 : SUCCESS in 41m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32c1aa0eccd74abf89f48a9347941f89 : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d16d2c92f0d403c8c64eabe54de857b : SUCCESS in 1h 55m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b58c2bd5f0f489aad7ebb2418607f23 : FAILURE in 2h 34m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db47aec4b8c0453cb24ff0d425193125 : SUCCESS in 2h 07m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/58fa7ca872dc443aab82437736797450 : FAILURE in 3h 12m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/45d9fa904f704864b3a8c2c155f40bb1 : SUCCESS in 2h 33m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c359afb32bbd4fe789f55a4165f4143e : SUCCESS in 38m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f0f8581ef1394021b8b732a565445abb : SUCCESS in 1h 20m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ffe2cbc044a4741843b3309297ee88d : SUCCESS in 1h 48m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2b75c28a428499c96b9a61fdc93fe21 : SUCCESS in 45m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ae53d22e319471e9b431451003d02a1 : SUCCESS in 1h 41m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0368ff3e5581424782126f42d43a50b0 : SUCCESS in 51m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"a046d71532578d8a4b579c42dc8225b4a92fe921","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-04 01:14:26.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-8-fc : FAILURE in 10m 25s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-8-iscsi : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":8},{"id":"23ebc505b69cd9dbd9cdf6fbaa443ea1c5119aca","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-08 15:29:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/80/968380/8/20260708183143/NEC-ISCSI : SUCCESS in 5h 34m 19s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/80/968380/8/20260708183143/NEC-FC : SUCCESS in 5h 58m 02s","accounts_in_message":[],"_revision_number":8},{"id":"7f72b943543797e0fa00ef81e23cf3b11b703dba","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-09 09:24:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4228/.","accounts_in_message":[],"_revision_number":7},{"id":"9a0794239925e345f2f7e00b6b2bcabd766005bf","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-10 14:47:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"20a6666126ea99c1f438298132f5b1f3460a08ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-10 19:06:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"29bafa6ff80293de6dab876b38001812d52eb931","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-10 19:08:24.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"73c1d6fb978f48e50d6fa9b3c32273753bc6d001","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 19:46:27.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/99aacce8381e4e59b9e3d4dc76b07621/","accounts_in_message":[],"_revision_number":9},{"id":"29f223ccade534a788a26accf506dee4c4759843","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 19:51:51.000000000","message":"Patch Set 9:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/579 : SUCCESS in 0h 44m 47s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/968380","accounts_in_message":[],"_revision_number":9},{"id":"7b29c0a97037e3e930976b3201b6e1abf6e85f09","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 19:52:00.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/849 : SUCCESS in 0h 44m 52s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/968380","accounts_in_message":[],"_revision_number":9},{"id":"53938aca898296a620684812f16b1fbdf42ba7a3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 19:52:13.000000000","message":"Patch Set 9:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/493 : SUCCESS in 0h 45m 10s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/968380","accounts_in_message":[],"_revision_number":9},{"id":"aab1b9a89e362c78a3c3cdb9f02ff0360cd3fa45","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 20:03:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/99aacce8381e4e59b9e3d4dc76b07621\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5b57cf7bdf042ada34dcbeb710a4b31 : SUCCESS in 56m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9b21d7e854844f4929907a888fd29c5 : SUCCESS in 40m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d06a61caac1545439053848a35c3bc94 : SUCCESS in 39m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e21f47a00fb74f37b38dfca052789413 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":9},{"id":"bc1ac8e670b05d65bd953fb4ea5c87ed40b589ed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-10 20:08:06.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":9},{"id":"89b3ab34bdd2789a31173c03ee06df470b7dc6d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 21:35:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fabf09730314f6e9ca87a63c337feb8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ee11121864145eea37a3bc7ef552237 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95a6bfee0e8e4967846a715a1defe8a7 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a98c19001ece41dfacd636f37d68bb65 : SUCCESS in 26m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cec2a38283dfeb96de2e7401de0051648f66ad75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-10 22:14:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/16bf72f6b7004a46875daeee41c6db17 : SUCCESS in 51m 25s\n- cinder-storpool-tempest-iscsi-multipath build/d8ba954977cf4de7867931c47140a406 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":9},{"id":"ae709f7c89fb4d535a4c2667f3fe03c798bc4d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 22:19:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb40fa4bfa84dc5a010bd5a34507db9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d48f73764e24ce8b92a9e6d7fa6fbee : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e93539e79f3f4a94ae15ec3bd0a24d40 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03c9ee84b15942a2ad01c64a5dab33d3 : SUCCESS in 7m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/deae8d6b0486469388dd0a289cc96e55 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057669e20d694ab2b6aa0a79bc930ade : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/10260b4c4a0d43d6a650d8f409ed59b9 : SUCCESS in 53m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/973e3a16d1bd42bcbaa71272b0120167 : SUCCESS in 1h 18m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c97b84acab874cf18c2a5372680fdfb4 : SUCCESS in 1h 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f6826c8c8834b289ccddcf0f55249cb : SUCCESS in 57m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17761e7f6b494e1cb6c822e010ec9a43 : SUCCESS in 8m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1fd582a93a3f496fac4f66467f915024 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f211d5ba5b294d9fac48ecc034fbf2ee : SUCCESS in 4m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91b7bed234ff46a38fb9abcd531a63f8 : SUCCESS in 7m 00s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3d75950e810240faa052dfc06e79136b : FAILURE in 13m 22s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e55ef1bc36f248099cd8dd6e8975b2f6 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2df4e79fec045628b9721cf3a91221c : SUCCESS in 6m 17s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8e4be3178f6741488b70c23ac20b8522 : SUCCESS in 9m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/836ab13ce4364c499da343d756104db2 : SUCCESS in 43m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5658b38e057437282651e8d0515ba5f : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/40dacabcebb740c9b79390b0c361d30d : SUCCESS in 2h 05m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0031a31254934d9591aa929d8f7e6d8b : FAILURE in 2h 31m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1410e2191ad4cfbbed14901bae5699a : SUCCESS in 2h 01m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e115f332c84149e9872e21194a694f7d : FAILURE in 2h 57m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/af2d9283b85f4b4b9b36b8b73faa8da2 : SUCCESS in 1h 59m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/656573ea9b1e4466aea97be787fd8dc5 : SUCCESS in 17m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/84d8f12ba1014431bbb1801c0dc0d477 : SUCCESS in 1h 16m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/681bb1ea46474de19e9ba2e42f69ed05 : SUCCESS in 1h 48m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/80a8617bca414c21ae538b378fe6f310 : SUCCESS in 32m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e234c812c8d44d1ca87a9107f9a74cd6 : SUCCESS in 54m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aee9170b60e24d72917fe171279eb7dc : SUCCESS in 58m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"92c39b98ae1b21eb01df9ba78592a2efc10d367f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-10 22:29:09.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_07_2026_22_22_56\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"b04f6e554b802280b62d3c0783303560ee7bef6a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-11 01:22:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4771/.","accounts_in_message":[],"_revision_number":9},{"id":"4b9b5d2d4951c2fcaabe3adf9728e0f1c7c96e7a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-11 13:57:34.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-968380-9/FC/ : FAILURE in 18h 50m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-968380-9/iSCSI/ : SUCCESS in 2h 30m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"f456493df98347a32b8735d307d7888d6c49b675","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-11 15:41:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-9-fc : SUCCESS in 59m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-9-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"8ff82af44617df8191a1d5b09c2bf47e4244223e","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-13 20:29:25.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"62c6abcd92ea23e2c981ca3764f5a855b6b14b85","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 15:23:45.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"b41f0b81d197526eef60f6d97d3c5719ee623f14","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-15 15:44:05.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a56ef3b1a61df73d5b160a38d07d7b6150da2a76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 18:47:26.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4442/.","accounts_in_message":[],"_revision_number":8},{"id":"e0158bd93300d3020de942ea8268656f7540135b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 00:53:46.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4595/.","accounts_in_message":[],"_revision_number":9},{"id":"592080d97c3ce6852009144cfe06b4a28d02927e","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-17 05:48:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"50a7286f0046fd6d62fd7199f5fdec464b4b8cc4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-18 19:40:26.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c2f46fa1303af97d8d1a0d2a548d32975fecab66","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-18 19:41:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"09586620c74762048e28978fd9159241e2e272c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-19 08:08:59.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"0b2a9f447fc8d3aace5c8df90b221340797722a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-19 08:19:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"31a0709cb2e54f64354c45bfa1552bc2c0450d9f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-19 08:21:12.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"52cff5f4e07ce42b05f82a14f7a51b73f60b5893","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 08:49:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d37ab07a344e31b1afd4ad3fbfeba9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/546bd78f4bed4122aa96fdda5b807a97 : FAILURE in 7m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4929a1d9e3684117afa7aff309ebabfb : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b78de435e7d417395ad4bdd409312f9 : SUCCESS in 27m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"97afc8517e1d77e47f8e7e1504d51a185e27e014","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 08:56:10.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/331dd0aa96a249fe87a15c0b8ad715da/","accounts_in_message":[],"_revision_number":11},{"id":"382d1241d161198c1fa8175af11de6736248cdfb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 09:07:24.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":10},{"id":"8241c7d1cd3cde510df5a4fdc6b3d6ed87ea7998","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 09:15:33.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":11},{"id":"e6a8ca0eab430a4516ea0f8b94e866cf9d83ab9f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-19 09:26:28.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/331dd0aa96a249fe87a15c0b8ad715da\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86f36bcb8c7847ba9b8e1b1eaa357d95 : SUCCESS in 35m 58s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f556acd2fa07439f86f1d2c1ba5707d2 : SUCCESS in 30m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c9481154cb0944bebba6c5a7706782a4 : SUCCESS in 41m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a31a437528b3407eb07168e0b7baa51c : SUCCESS in 30m 51s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/90d607e922114168b3cd00bd3e2bd91a : FAILURE in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7525e5f5f28072cb5fc9719eae0fc305ba5f9ddf","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-19 09:34:01.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/983 : SUCCESS in 1h 11m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/968380","accounts_in_message":[],"_revision_number":11},{"id":"2f12a8a1449049d67f147d088703c0726896925a","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-19 09:36:19.000000000","message":"Patch Set 11:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/889 : SUCCESS in 1h 13m 51s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/968380","accounts_in_message":[],"_revision_number":11},{"id":"b0db65bbc345e8240a53495d40a6c2181cd82749","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-19 09:39:05.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1252 : SUCCESS in 1h 16m 35s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/968380","accounts_in_message":[],"_revision_number":11},{"id":"2a5b3e016d724ceefea3e486895d075db6f5b8ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-19 10:31:22.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8b1cff87fa0e4cfdb61ade07cb54b5c5 : RETRY_LIMIT in 48m 55s\n- cinder-storpool-tempest-iscsi-multipath build/a18991a019f34b1ca4c9467df8da7a98 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":11},{"id":"ec64cff8729f12fda36ff78fb1978de461342d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 10:57:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b714a8c4929e4ed282bb5a16a3137ff2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e01f5dff694e3ea02290fd2455ff6a : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/146948f3ff604a3d9c6d57537f31c859 : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/053c1499e48846d5bc803bc3ab74074a : SUCCESS in 8m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf2955ff8ff547b7a256536b4fad930a : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26eaa9dea3eb4680ac388445dec9947f : SUCCESS in 13m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/84c37a94c61241849adb8857b4aed07f : SUCCESS in 1h 10m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3523768eab63495183246c0558e2fa36 : SUCCESS in 1h 05m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc141310a8f549e6a1c2c272ce6b337d : SUCCESS in 1h 32m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6231d2b8cea247aabaa1bb4be92a208e : SUCCESS in 1h 01m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bc61ebfa47348cca79751ed9e6e6e64 : SUCCESS in 9m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18208e9e77664d2aaa8cbc8db609d332 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/46591e9cf2b847d0adead1260779c17f : SUCCESS in 4m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4ca6cd2465f84fba8638e8504d014f85 : SUCCESS in 6m 48s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/ddd64d96e4964e80b00c5bdc5468f17d : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bba13f0e02994916b26696f1fee97c6b : SUCCESS in 7m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5a45564810a443aacaf6694b54ee74b : SUCCESS in 6m 17s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7df900d5ee0044f39ccafc6b7973f881 : SUCCESS in 10m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0dd03d20a2a14ae8826d35f94d757194 : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/196e145fe9044de285a3d2a5d4363b08 : SUCCESS in 3m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/543b2fde91a04181a0ac9f5fb931fad9 : SUCCESS in 1h 27m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/75795cdb4b874c0a96e0eca9e10611d6 : FAILURE in 42m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14e13e580b224583b4a15c8b3c5334cd : SUCCESS in 1h 24m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/45d2804a656840d68965e5342eca4b4a : FAILURE in 2h 31m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4a64bfa3f47a4cf49a10a1a9a3a7ac8d : SUCCESS in 2h 03m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/46428dfce4764d4a9b0db1ba9a225190 : SUCCESS in 25m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e1ec203ab9f4c00a5ca74531db54bdf : SUCCESS in 52m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/979ded2b43f444468dff9bd3e763e20c : SUCCESS in 1h 25m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/76a2161473a14122af9763b226dcdf7f : SUCCESS in 40m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e545735bd24c4ff69bbff087c9f6d13a : SUCCESS in 1h 31m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5781e27338cf4889a47d7021ccead3a0 : SUCCESS in 29m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"1342bf6676cdb0496f1a2a13aae9cebcba1d8ed9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-19 11:08:06.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_08_2026_11_01_14\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"e37a8d2f2fd1fd0aaf54b3de1ef1c35b864b259e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-19 13:30:23.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1255 : SUCCESS in 0h 46m 35s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/968380","accounts_in_message":[],"_revision_number":10},{"id":"6e1c1bb776f05e50fc776bdbbd6162b46d9a6dd1","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-19 13:31:39.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/893 : SUCCESS in 0h 48m 15s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/968380","accounts_in_message":[],"_revision_number":10},{"id":"ffd6e69696ad87e8fff6bb8ccde1676f793ee584","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-20 02:24:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-11-fc : FAILURE in 10m 23s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-11-iscsi : FAILURE in 1h 22m 46s","accounts_in_message":[],"_revision_number":11},{"id":"3ed9cb74d92c6374f5f3dc09e279717647f6aafe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 13:25:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5197/.","accounts_in_message":[],"_revision_number":10},{"id":"d887d0fb03bf7e5d4c9ac0a912d609709247ba18","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 15:00:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5021/.","accounts_in_message":[],"_revision_number":10},{"id":"85f1a71d6e7c1e143bac55d08ef28972ea11526f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 15:23:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5198/.","accounts_in_message":[],"_revision_number":11},{"id":"b88f8db8aad36d12060703753d38bc287dcfa3a7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-21 16:56:19.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5022/.","accounts_in_message":[],"_revision_number":11},{"id":"188b4e965f78a59242223d27049c49472dc8bb6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-23 12:59:25.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ffd53cd88f769ae464845579fee082a55cd392a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-23 13:16:31.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0e3d1f356ff14ac8984f150236b6b723/","accounts_in_message":[],"_revision_number":12},{"id":"2f291a3f5ec30ad5f024ee96e71aeeda20308735","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 13:26:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c86f54fb5cc84b61bac6caea3de6453e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/95d87fcb57a6488dae4003b8f85c4826 : NODE_FAILURE Node(set) request 442ebdc6a73547888a5ccd80f34c6bd8 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/447add43fc9945e8b5e6875dc318a2e1 : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4154a460d264278aea7d42e256a4b86 : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ebb12be4d7655a0f1beed86d0ae1845147923fa2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-23 13:40:57.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0e3d1f356ff14ac8984f150236b6b723\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/416a1d76d43a42c4a14f2888182eae77 : SUCCESS in 35m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/75e2d7396da549fb89eb154b3aa898b1 : SUCCESS in 31m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8ea27a293394b5ebc782f03b9725b9e : SUCCESS in 40m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f931957889e249b6b25f4b66a8c9fcc4 : SUCCESS in 33m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7615408170734f39accaface1ff90a22 : FAILURE in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dd6370b5ece71a809b1a47e8762637ea55258595","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-23 13:47:59.000000000","message":"Patch Set 12:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1272 : SUCCESS in 0h 48m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/968380","accounts_in_message":[],"_revision_number":12},{"id":"11892d531b3277a3f44196a3ac1d8d105015508e","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-23 13:48:03.000000000","message":"Patch Set 12:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/910 : SUCCESS in 0h 48m 13s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/968380","accounts_in_message":[],"_revision_number":12},{"id":"e4022b064cc719aa65e6874263f53607ae55b636","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-23 13:48:15.000000000","message":"Patch Set 12:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1003 : SUCCESS in 0h 48m 24s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/968380","accounts_in_message":[],"_revision_number":12},{"id":"69a1e84154cddf3dd1107f3cf70d00098a7b0167","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-23 13:53:49.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":12},{"id":"543cf96392f9e036e7fc66ab766774d14404d82c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-23 13:54:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e5c1da72b06b4dd0aead6b64ee257f18 : SUCCESS in 49m 04s\n- cinder-storpool-tempest-iscsi-multipath build/4b1edf924c424c9fad2a08b0681a5aea : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":12},{"id":"8716bc6e76090f13f878223f95679fe3bef12721","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-23 14:54:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5221/.","accounts_in_message":[],"_revision_number":12},{"id":"7d269244caf72ff15db071c488231e120d1c9db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 15:16:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc971f0e898d4019b1752c69b7b293af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7849f81c6d6d4a30b4b88051bb20b257 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51803e0c181a4397ba3929dcc1d5336e : SUCCESS in 9m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0f7ebaed62d4e83abd170eadc8e5aa2 : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/915f08c349994ac79907e07d7473a440 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f0d920d14e40c1a33277e1f73d8b89 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/c280639ddae245709037ad7bec5dca06 : SUCCESS in 59m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3ec678c0c2b4fc8a8899e824edeb583 : SUCCESS in 32m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/313cd7c658c949799104eadc19875c00 : SUCCESS in 1h 21m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b9f4cf6ff1e4c9ca2e527e321dfd0c2 : SUCCESS in 37m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/366a4316366542d982269d497f96c290 : SUCCESS in 8m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/05321caafe964b1dba9e82055ba1f8c2 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2fd289f633f04ed4a719c1b40ce74d0b : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/06d9f6d0597f47ac805f60bec8f05ab2 : SUCCESS in 4m 41s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f20d2eccee514ff498ac0d84f37ce6b4 : FAILURE in 6m 52s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/16016be038f740d3934252deee72529a : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef8d15d3384a4c9d83216be71c6b6af4 : SUCCESS in 6m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/347d8526447741f78a2264fd6d210c54 : SUCCESS in 6m 05s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a29132be0b2b4eeaa22341f85463e159 : FAILURE in 4m 35s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/023b45a3f75b4826b8b15e3afc5f07ae : SUCCESS in 7m 01s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b86cb2ab01344062a9a90614df29ed4b : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97ea89b85e6543edaa9afb346709c429 : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/42cd2ddc75fa441d8aa7f6fe212db91e : SUCCESS in 56m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ec190535e474718b9494c6167008abf : SUCCESS in 2h 06m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ec58d6ebc52e4b93af0efa6df44b3c34 : SUCCESS in 1h 40m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b93a26d5687140b7bb7c9de9b28f3aa6 : FAILURE in 1h 55m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/baba1b92602f42b4981c542a4994eb7d : SUCCESS in 1h 23m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/11f30939f3bd48808c0bfd157262942b : SUCCESS in 23m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90eac5832f6549cbb3d8a6bfbdf3f4ed : SUCCESS in 1h 11m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6285dcde92e8488197a176f4c6e172b2 : SUCCESS in 1h 56m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/432f9e8566e44824968ad68f6589ff88 : SUCCESS in 33m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5efc7f8663b54b8a9784762f2ec8a605 : SUCCESS in 1h 07m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf72512f6df746fe999e9d8b2f7fcc00 : SUCCESS in 57m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"07680a4063062e9409158b19a4131efa06faee89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-23 15:26:56.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_08_2026_15_19_59\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"2e50c083e8f4efdfa5028fca1929fdec730063ac","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-23 15:41:59.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-12-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-12-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"65176d93d4c447a2c4b37161c04e539f7d7564b5","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-26 16:53:34.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8d42ca410500373ed2e77748cf520d9605afa400","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-28 14:26:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7332d52d98deb79acd8a46c651b6515c282552a8","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-28 14:27:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c151fd8c57f60a6e8655a1d3a0132ec1b45e078b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-02 23:04:51.000000000","message":"Patch Set 12: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"cb4a983fff784ea8308f62108d9894c4d0270061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-09-04 12:01:23.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"960e7450288bf450cf71dfe45a61bd4c26847b5c","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-09-04 12:12:43.000000000","message":"Patch Set 13:\n\n(10 comments)","accounts_in_message":[],"_revision_number":13},{"id":"8f247b1f0d295ee7b583cc6e2d7ff7278961d239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 12:31:01.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5cba52f8c3746569c067a9e2d647674\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8fa4aa127554e3286dd32e28ec1b7d1 : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3208405ae1124bbdb3b2953c1bdd2741 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01127ff264cd411b8ff6b73a527aaed9 : SUCCESS in 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"796899596b1e24244f3a6dc08901ddf4086f760c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 12:35:43.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4b41bfd9cf5445a6aa03804e415068d2/","accounts_in_message":[],"_revision_number":13},{"id":"d966e0f3531e0dc254e7587981e2818ca8351538","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 12:56:18.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2c5df478acd1f5840ad8d572904144448e127f46 .","accounts_in_message":[],"_revision_number":13},{"id":"5f19cdd773453e8770741b24904c15092a49ca6e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-04 12:58:56.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/56bc4cbed9dc4e61a10d14b31fd14ae4 : RETRY_LIMIT in 16m 50s\n- cinder-storpool-tempest-iscsi-multipath build/db16a64bef82486b8d0b229161946971 : RETRY_LIMIT in 18m 40s","accounts_in_message":[],"_revision_number":13},{"id":"c4123450c89ebe3b1c172fb64f947b1b81b17def","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 13:04:17.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4b41bfd9cf5445a6aa03804e415068d2\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/44874bf1d873423394ffe95f61a36513 : SUCCESS in 41m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c15df5f6e18d407aa41e393bce395cf2 : SUCCESS in 39m 04s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71cbc96b679844f781576343f49fe366 : SUCCESS in 46m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/080bfedebe2f4d2897fd215a83f37566 : SUCCESS in 30m 55s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c6bc097585543dea45c468563d9e81e : FAILURE in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"07179ba3c029be6c2f2744804f368b62eeaf8a05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 14:07:10.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5411/.","accounts_in_message":[],"_revision_number":13},{"id":"afa9e656de290e7fe6cce2ad2296b32354b5df67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 14:38:27.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d22ac520ee1471589e24447f041da8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de450ec8a3394b5d917db76ef2ed8399 : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db68700b075a49c69633082bdf1efdd8 : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e72dffe739a41bd91288275cb286b42 : TIMED_OUT in 40m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6750749b6c14a7aba52b157e0f3f6c4 : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59675dc066da486398ae9615b34e218a : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5b85a55b2fbc4d45803d723d8c456b89 : SUCCESS in 1h 10m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c79a94999cd4da5b26ff78895ad277b : SUCCESS in 57m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06df930baff24c60831b13de59382218 : SUCCESS in 1h 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96020e64cbc34561ab317d7c7f9fdbe7 : SUCCESS in 1h 02m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777b71f0053d4c12853a1b720716a715 : SUCCESS in 10m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/14c45ea7c7384d088b138adc8c995d65 : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e28998f4ec404a20ab9e907474a1d0d0 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bde669ab457448c2a442af385cd1364e : SUCCESS in 6m 24s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/6ff03e008fd7466eb37549ccff17ee79 : FAILURE in 11m 30s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2df0a49d44614138ac0c6416efaadb6a : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/33200a04e6af446188c6a7d457badac6 : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/52ad8db7ecdf4893886d47d0e2e9ee05 : SUCCESS in 5m 49s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3935545f110c424d8f3c23dcb7ad2564 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4f7923d95f8544c0bbaddaf0c9269c48 : SUCCESS in 9m 10s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9736e5f37ba84a8d8302ec3c4a4c4eb4 : SUCCESS in 37m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c0b5608053646a7950e91506efe37b1 : FAILURE in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c18b89995062485da5f30101e5bdd66a : SUCCESS in 1h 43m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ae1a25c2cdb4828abee9f74ec3959de : FAILURE in 1h 59m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6c9d2ce0cd54f8495ae548d8d949c2d : SUCCESS in 1h 48m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9ec28e4ae94a4bbbb23579c29f52dc52 : FAILURE in 2h 31m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f97a5336e9bb42958529a2fb075c8213 : SUCCESS in 1h 37m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fd3a5b830d1f4c6590eed263dac434bf : SUCCESS in 33m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f7a87ba7e534cad99ea2ac5930860d3 : SUCCESS in 1h 16m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/89bb4651e88d4863b708106cd6aa1180 : SUCCESS in 1h 27m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4f787e6c57ce453bb0e5a3ce8b7b4ff0 : SUCCESS in 29m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c32322fe3f3047f0b97ddf97c982c865 : SUCCESS in 1h 32m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b751d4c0f7544acbf6a6563c7c641e1 : SUCCESS in 1h 02m 52s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"ac8e08cdb1fef8cdaa31833d5e565a23faf7c6f3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-05 01:42:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"52f1554c645ae7dd08766d3e6b2c5e4988e7281c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-05 01:46:48.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8f294ddf0cc565c4c143ea0f7f014da4bede9ab8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-05 02:05:12.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-13-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"aec708a03c1c4141565c19177e0f49547af63d1d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-05 02:27:40.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/27e71c7b28884850a85f2b2cead9ffe9 : RETRY_LIMIT in 43m 38s\n- cinder-storpool-tempest-iscsi-multipath build/6adf16179734487abf7107e0f0248bc6 : RETRY_LIMIT in 35m 52s","accounts_in_message":[],"_revision_number":13},{"id":"2456820ea3c3e296f140c7fbe458f4ae76c65cb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 02:45:49.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54de6b0c1c334ffc891533689f0247d2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39969f0e05ac49769aa41ee9c6b0537a : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a73639f8a4af43bd9bb27d9af876e68f : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0efff75a64eb43bca4419f99ba9e4cd6 : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"abcfeb81bda3964ce4f389337a3dedb9270e70d6","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-05 05:35:58.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-13-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-968380-13-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"94cfb8cc4f58bc030d37b543519807f4824386dd","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-09-07 04:22:32.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a9f4b6828f0921d83c4bfb7e61d9ea4767a83dad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 04:40:24.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/57edf38db552460b94e1e67f90993d00/","accounts_in_message":[],"_revision_number":13},{"id":"ec6d60b229790add29a7b67c5c55d1e961851782","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 04:42:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d958064954804747a50c69b87dd5f999 : RETRY_LIMIT in 17m 04s\n- cinder-storpool-tempest-iscsi-multipath build/e25ef08697c94cf5a280edcb430d118d : RETRY_LIMIT in 17m 59s","accounts_in_message":[],"_revision_number":13},{"id":"c5783fa31d8423ad784381f18a891b46df0fc733","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 04:48:13.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5c914e087440fba0b092c4e5fe1398\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a277cbcfa15495792cadb3d0af59792 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff3195ddf5d14961a4c6a14e50be510d : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3e778814055a4aa089fe0664cd684d9d : SUCCESS in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7f094c623c3b213ffd8142a51cdd6b34b3df5f13","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 05:17:19.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/57edf38db552460b94e1e67f90993d00\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb77781e613d4fdf84a755a8bce630d1 : SUCCESS in 41m 14s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/62a1bccb1b2a4c4dbf4747012e21e2df : SUCCESS in 37m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d4eb06aa8774f8db12039e5b6da000d : SUCCESS in 54m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9bf1a06d4b114622bbfd8b2ceff951dd : SUCCESS in 36m 06s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f250d4376275409fa62d070ca37a808f : FAILURE in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8ebe5b58023cef68d4890a348c70039bc97248c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 06:47:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2fa3e63d5f4acf9873d014684f569e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9726c6533f254becbb8063ab2dc4b4a5 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c07a3a470cb4c7bb98344553ee7b154 : SUCCESS in 8m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d49a8b919464bcb84a1a7924461aa06 : SUCCESS in 11m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8e05319909d4901887f1577f3ce2f19 : TIMED_OUT in 40m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b213b6a2f0f4de9886291de25850007 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/e94f8f74c65f4e9cb72aa2cfd2746268 : SUCCESS in 29m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d1ad1e9d36c44aab56914bd8667fe54 : SUCCESS in 34m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/babee7cca02042bbb6328152433123e0 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7225fb805dd140ed8e3ae141a4ebc9e7 : SUCCESS in 56m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e37c81f83fe4747bb4eaa2a211a8a52 : SUCCESS in 6m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/121176edf201444981ba983cd980877e : SUCCESS in 20m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f900634032e041c18f54821f0a9de92c : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91637292a33a471ca691e7c015ab174c : SUCCESS in 6m 22s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8d1382d6b2fc44c58685c307bf3cd733 : SUCCESS in 11m 46s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/52ba621e70124361b6f66ecf3b9bbd9c : FAILURE in 6m 18s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/75d82ed176624e8896b344482a468c3c : SUCCESS in 5m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aba4d74094d64f509b67be2d6370b57e : SUCCESS in 5m 57s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f56a95395a0549ccaf483517970d83f7 : FAILURE in 4m 25s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/febfe3553282496790dc5ecee60d699d : SUCCESS in 6m 04s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1de665fa40764a0db0650510943ee240 : SUCCESS in 38m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/008ecfc58ffe4630ba272398e72476cf : FAILURE in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d47f7b7b44ca4193a91f6d0d9ac7d728 : SUCCESS in 1h 31m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc6b385229d642b29829a9562499b4bb : FAILURE in 2h 08m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/445d02746ef34d16868352880fcfa4c7 : SUCCESS in 1h 44m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/872f9c8988484ea895149b0996358707 : FAILURE in 2h 19m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/185c39a7550347809c052d81b3e2a1cc : SUCCESS in 1h 29m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/530de60780424b6586353a767d03add9 : SUCCESS in 33m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7de89ea1b0c4618a2b6723ce2b7063f : SUCCESS in 1h 14m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2f41dbc7556e4869a79d69beba7ec798 : SUCCESS in 1h 13m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/85fd070230bd48e997119d2d6b9e7746 : SUCCESS in 43m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b19db6e121c3414daa186f9b0514f7f2 : SUCCESS in 1h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39106d3c358241d284e63380e8190795 : SUCCESS in 40m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"720961b784e29dc6f890d06909c9d669231288ea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 06:57:17.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_06_51_42\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"6fafef4f7d5c8e26a4335205966a5dc1cf4b58fb","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-09 05:17:45.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7883008ab7ab77f9f9d17f5e2f923c788f8b944e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 05:18:34.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"6197408197430160c77ab3e53e9726d40b46cdc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 06:54:14.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c353603e1ae49239be7a8c46f007b60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64df1681532d409e800b342a0c745762 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1b8315e0b9d46f9bca32070d2a2ee29 : SUCCESS in 14m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b499e18681443bf90d5eddb8cbd35a9 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceca387de60443cf973681b130076a7b : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f11d22819b5e4fcb8389862fc20d3049 : SUCCESS in 1h 06m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3724ec1c27bd4e36b852904f9fbf29eb : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bacaa5279f149538d673b0965624c65 : SUCCESS in 1h 30m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb22c33616a348fc915f2af0a688a2ca : SUCCESS in 57m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2530f808957468486ccb7acf09172b3 : SUCCESS in 9m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcbcb62161244a25bc8cbc1d772af9b7 : SUCCESS in 1h 14m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d23eb62ee20c480790b8d7ac3abcac7f : SUCCESS in 1h 26m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee8a9cf21eff42898657ab72f3b65d8b : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":13},{"id":"368848510d4ade7b53c9c77c85d66f8ad6e2a91d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 06:54:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"dc4127c1a6015cfc148a12f6df930d2805348d4a","revisions":{"c7f799206cf55757683ba22fe7d35f67c719b33e":{"kind":"REWORK","_number":1,"created":"2025-11-25 19:39:03.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/1"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:37:29.000000000","tz":-300},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7f799206cf55757683ba22fe7d35f67c719b33e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7f799206cf55757683ba22fe7d35f67c719b33e"}]},"branch":"refs/heads/master"},"6f4dee5ca4a6f8823131edbb0538b0614aab3a48":{"kind":"REWORK","_number":2,"created":"2025-11-26 04:03:48.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/2"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-26 04:03:06.000000000","tz":-300},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f4dee5ca4a6f8823131edbb0538b0614aab3a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f4dee5ca4a6f8823131edbb0538b0614aab3a48"}]},"branch":"refs/heads/master"},"33b68b8027488402ac47338b29979099fec03634":{"kind":"REWORK","_number":3,"created":"2025-11-26 04:05:33.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/3"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-26 04:05:21.000000000","tz":-300},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33b68b8027488402ac47338b29979099fec03634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33b68b8027488402ac47338b29979099fec03634"}]},"branch":"refs/heads/master"},"9d6c7806cf3ae2322d1b424d33bd6b609edb3815":{"kind":"REWORK","_number":4,"created":"2026-01-06 13:09:52.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/4"}}},"commit":{"parents":[{"commit":"00505d25879992b268402a71125ccc65b3c4088d","subject":"Merge \"mypy: NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/00505d25879992b268402a71125ccc65b3c4088d"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-01-06 13:08:41.000000000","tz":-300},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d6c7806cf3ae2322d1b424d33bd6b609edb3815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d6c7806cf3ae2322d1b424d33bd6b609edb3815"}]},"branch":"refs/heads/master"},"c173b5b41e756db0f6bd1e8397e941e7e84cc756":{"kind":"REWORK","_number":5,"created":"2026-04-13 14:57:16.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/5"}}},"commit":{"parents":[{"commit":"a707d2b4db22c346deb1d1b16c0607a43aac65d8","subject":"Merge \"NetApp ONTAP:Add async single CG replication support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a707d2b4db22c346deb1d1b16c0607a43aac65d8"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-04-13 14:56:45.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c173b5b41e756db0f6bd1e8397e941e7e84cc756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c173b5b41e756db0f6bd1e8397e941e7e84cc756"}]},"branch":"refs/heads/master"},"ae7b5b4cc9480f16d42e5a7f5b10794b84b20693":{"kind":"REWORK","_number":6,"created":"2026-04-16 13:41:27.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/6"}}},"commit":{"parents":[{"commit":"b1015d9cb2aabf07c472817c58a2957f50671b0d","subject":"Merge \"Re-enable flake8-import-order for tox pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1015d9cb2aabf07c472817c58a2957f50671b0d"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-04-16 13:41:02.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae7b5b4cc9480f16d42e5a7f5b10794b84b20693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ae7b5b4cc9480f16d42e5a7f5b10794b84b20693"}]},"branch":"refs/heads/master"},"11eb8295fab5f71028254db49b7f5463a4dbc70d":{"kind":"REWORK","_number":7,"created":"2026-06-09 05:02:04.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/7"}}},"commit":{"parents":[{"commit":"9524108fd6a902e3c3866cb4943bf2a7f0877ce8","subject":"Merge \"db: Use cinder.db.api directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9524108fd6a902e3c3866cb4943bf2a7f0877ce8"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-06-06 12:50:19.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11eb8295fab5f71028254db49b7f5463a4dbc70d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11eb8295fab5f71028254db49b7f5463a4dbc70d"}]},"branch":"refs/heads/master"},"f452a0af73dfea32709f78a6c22c807abc311b4b":{"kind":"REWORK","_number":8,"created":"2026-07-01 16:10:25.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/8"}}},"commit":{"parents":[{"commit":"9524108fd6a902e3c3866cb4943bf2a7f0877ce8","subject":"Merge \"db: Use cinder.db.api directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9524108fd6a902e3c3866cb4943bf2a7f0877ce8"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-07-01 16:07:14.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f452a0af73dfea32709f78a6c22c807abc311b4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f452a0af73dfea32709f78a6c22c807abc311b4b"}]},"branch":"refs/heads/master"},"9ea8f32d931aef65ee2328e7c16784274f70b9c1":{"kind":"REWORK","_number":9,"created":"2026-07-10 19:06:44.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/9"}}},"commit":{"parents":[{"commit":"9524108fd6a902e3c3866cb4943bf2a7f0877ce8","subject":"Merge \"db: Use cinder.db.api directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9524108fd6a902e3c3866cb4943bf2a7f0877ce8"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-07-10 19:03:28.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ea8f32d931aef65ee2328e7c16784274f70b9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ea8f32d931aef65ee2328e7c16784274f70b9c1"}]},"branch":"refs/heads/master"},"fc9b861ce34dc01e18ece629602ff35bd83f44c5":{"kind":"REWORK","_number":10,"created":"2026-08-19 08:08:59.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/10"}}},"commit":{"parents":[{"commit":"2a29bacf6c78cdb55484ac45a7626d071ea3119b","subject":"privsep/nvmet: support both Port.setup signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a29bacf6c78cdb55484ac45a7626d071ea3119b"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-08-19 08:08:42.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc9b861ce34dc01e18ece629602ff35bd83f44c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc9b861ce34dc01e18ece629602ff35bd83f44c5"}]},"branch":"refs/heads/master"},"18e20a7a14cf120ec8098c5447651c7cdaedd9b9":{"kind":"REWORK","_number":11,"created":"2026-08-19 08:19:04.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/11"}}},"commit":{"parents":[{"commit":"2a29bacf6c78cdb55484ac45a7626d071ea3119b","subject":"privsep/nvmet: support both Port.setup signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a29bacf6c78cdb55484ac45a7626d071ea3119b"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-08-19 08:18:50.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18e20a7a14cf120ec8098c5447651c7cdaedd9b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18e20a7a14cf120ec8098c5447651c7cdaedd9b9"}]},"branch":"refs/heads/master"},"df215c4167a1f7fd6a5b5b8fb7c0e78505a63c45":{"kind":"REWORK","_number":12,"created":"2026-08-23 12:59:25.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/12"}}},"commit":{"parents":[{"commit":"ca4708870cf0516274263c6c45b44705bd593cc0","subject":"Merge \"NetApp:Fix to handle error during consistency group expand operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca4708870cf0516274263c6c45b44705bd593cc0"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-08-23 12:30:14.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df215c4167a1f7fd6a5b5b8fb7c0e78505a63c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df215c4167a1f7fd6a5b5b8fb7c0e78505a63c45"}]},"branch":"refs/heads/master"},"dc4127c1a6015cfc148a12f6df930d2805348d4a":{"kind":"REWORK","_number":13,"created":"2026-09-04 12:01:23.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/80/968380/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/968380/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/968380/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/968380/13"}}},"commit":{"parents":[{"commit":"c409487fb0e929e02e8e07e391c99d21025cf5b6","subject":"Merge \"NetApp ONTAP retype/migrate to a non-NetApp backend fails with host \u0027none\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c409487fb0e929e02e8e07e391c99d21025cf5b6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2025-11-25 19:05:06.000000000","tz":-300},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2026-09-04 12:01:07.000000000","tz":-240},"subject":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update","message":"NetApp NFS/NVME Driver – Optimizes the efficiency of pool stats update\n\nImplemented logic to update the performance and deduplication statistics\nof backend pools at configurable intervals for NFS and NVMe/TCP driver.\nUser can use below options to control the frequency of performance\nand deduplication data retrieval from ONTAP\nnetapp_performance_cache_expiry_duration and\nnetapp_dedupe_cache_expiry_duration.\n\nThese options are needed to fine-tune the frequency of querying\nperformance and deduplication metrics, as these queries occur both\nduring the delete volume workflow and in the update_volume_stats\nmethod, which is triggered by the periodic interval task. Users can\nconfigure these parameters to prevent the queries from running\nsimultaneously when both workflows regularly coincide.\n\nPerformance statistics updates are further optimized by caching the\nnode details to avoid the ONTAP call.\n\nCloses-Bug: #2132785\nChange-Id: I2c5df478acd1f5840ad8d572904144448e127f46\nSigned-off-by: agireesh \u003cgawasthi2010@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc4127c1a6015cfc148a12f6df930d2805348d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc4127c1a6015cfc148a12f6df930d2805348d4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
