)]}'
{"id":"openstack%2Fcinder~968831","triplet_id":"openstack%2Fcinder~master~I81e5198089b3278e84e45a37c7734c45c3104e4a","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-17 10:33:24.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2025-11-28 18:09:01.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2025-11-28 18:09:01.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-04-30 12:41:42.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-30 13:36:02.000000000","reason":"Someone else replied on the change"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-04-30 12:46:23.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-02-17 10:26:28.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-03-03 05:20:22.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-01-08 13:53:24.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I81e5198089b3278e84e45a37c7734c45c3104e4a","subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","status":"NEW","created":"2025-11-28 17:47:22.000000000","updated":"2026-06-19 19:57:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":57,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81b2c570620abb66849b14fdf99e0d3ccbbb53be","_number":968831,"virtual_id_number":968831,"owner":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-30 14:48:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 17:47:33.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-11-28 18:09:01.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-28 18:09:01.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2025-11-28 18:09:01.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-11-28 18:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 18:41:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-11-28 18:47:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 19:15:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 19:22:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-11-28 21:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-29 16:25:17.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-01 09:40:30.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2025-12-03 15:18:11.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2025-12-05 15:19:08.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2026-01-08 14:48:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-08 16:51:54.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-12 05:38:17.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-12 05:41:26.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-14 09:39:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-14 14:17:23.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-02-05 13:23:43.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2026-02-16 14:23:33.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-02-16 14:44:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-03-06 15:02:53.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-06 15:16:10.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-06 15:17:04.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-04-30 13:36:02.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-19 19:57:00.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"eb14784ccb1102abc6adb75a11fe1cfe0182edbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-28 17:47:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c9491029ea6c23eb71b7eff40826915fb05b6c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-28 17:47:33.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/82/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"70ba0b20375dcc5a4843e43f3e9f44bf46900e93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-28 17:48:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5d358e52d353428dcc57492f7c90ab768c574cb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-28 17:48:47.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/83/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9e84cc77580c441d1f5a04627f98ac9940ae27bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-28 18:03:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c474eaf702fdbbef015b461612972bcb01f4383","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-28 18:03:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/84/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0874618413aa644acd74aac2f3834d2d2d5cc5ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2025-11-28 18:07:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"484fe0162f303a7a5b783b4b8fef1877094e4309","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-11-28 18:07:36.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/85/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"6d852aa9bf5a7902061542c4fbc48edc931d29cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:32:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0a295264f544ccd8860a7257684a8a9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ca1825c01dc4c0296fa7aa2a90d7c6b : FAILURE in 15m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2a4b83330ba459da35c9db84222c51d : FAILURE in 23m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ee52d0b35244717945cd4a7bf28e15f : FAILURE in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ea4f25a9d3d8e5969f226a6dcdda6d85e38be3d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 18:41:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2586/.","accounts_in_message":[],"_revision_number":1},{"id":"a6d42a5a9ffd18b9a9cf684b7677d5a3e103f2eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 18:42:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2734/.","accounts_in_message":[],"_revision_number":1},{"id":"83a5620c623cf7fd57bf3f3751c764116f5f700f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-28 18:47:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":2},{"id":"982661e3e2481222b1dd72b0d55d8ba495167ba8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-28 18:48:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":1},{"id":"ee3326b3dc9d0576249a13d3c43b472f0a034189","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-28 19:02:03.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":3},{"id":"1fbb3e252c8857bb2aa513a1811a8ba5609bd10a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-11-28 19:02:13.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":4},{"id":"a297f18a666cf780ca964e8716bf5059c8f89128","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-11-28 19:15:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ed6801429e774485b55b50da54987c6c : SUCCESS in 55m 12s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e480b5a0e1124d719fa13c56ed12d860 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ecf481f15927706baa69b7d1311709c7bade5cbc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-11-28 19:22:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 1h 11m 16s","accounts_in_message":[],"_revision_number":4},{"id":"02e49852ef88c990c52e5d98620436471a2df74e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 19:34:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2735/.","accounts_in_message":[],"_revision_number":2},{"id":"4c8c88d3fc550bf77688f630f80526c422e0cebd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 19:38:48.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2587/.","accounts_in_message":[],"_revision_number":2},{"id":"77b32c5eced718010b0d28bd0085d4825636243e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 20:26:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2736/.","accounts_in_message":[],"_revision_number":3},{"id":"572bc50229bdc3550b962d9e87be18bb1813badf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 20:31:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2588/.","accounts_in_message":[],"_revision_number":3},{"id":"bdc90645965f2cedce22ef2d9cc8486aa9d28ea0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 21:18:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2737/.","accounts_in_message":[],"_revision_number":4},{"id":"8eb13e8a99a0bc4d56b8ae6f3ca0f99c37aed134","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-11-28 21:22:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2589/.","accounts_in_message":[],"_revision_number":4},{"id":"08434015b153ec69e7bf88882556f7c9f6668899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 21:24:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1cf43ec971749bca32f3bf4dd239a72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58900bcbfcc546a9a83c10f2a7f89a1c : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb7561b6ebb46a1985c4bac1ecc0cf9 : FAILURE in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53edeb6938734970af240a154847bdb5 : FAILURE in 14m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770fdf70fc854abfbac0109559620d52 : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/595cd18b02b243bdbf06e6ba196abaa0 : SUCCESS in 29m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f430dfbb96bb4907ba5c41c3cd6407bb : SUCCESS in 1h 08m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3af3156a52a24c489963c3977ecd4d0c : SUCCESS in 1h 40m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c4e607797804bd2bfeccc970b453196 : SUCCESS in 1h 01m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/86702247a17f45da93f4ff1f89786387 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76ead3999a1d44529fb352b53bcd7b6f : SUCCESS in 7m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09fe207d60ae4f4e9418251990535a90 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/068d6d20dfd2464bb1d690d365e55707 : SUCCESS in 4m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/248341b44bf14539ae59dbaa9078d39a : SUCCESS in 9m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c099ce2d2dad4bd6b7a94e4dd60baa90 : FAILURE in 45m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de2bc68a7f7944e2bffae15122f2e37a : SUCCESS in 7m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c7b34a58a7a4c85b5b72ae75ac2bdea : SUCCESS in 1h 56m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dca8d97032da4f2e990d5a9e08cfb2b1 : FAILURE in 2h 43m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6e7181fcd2e344958c7aa82e3389218d : SUCCESS in 2h 28m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2497c139a278402c9b57c432b1606775 : SUCCESS in 1h 41m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9fc66a4127d2417193022fa33d003491 : SUCCESS in 35m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89f3850fb27a427db56e3da6a11b114f : SUCCESS in 1h 15m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5dd0afd7588e4f4898a2244845e76ec4 : SUCCESS in 1h 35m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/782e5bf3b0d94575bf51a1e411bfdb57 : SUCCESS in 35m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d95382c3ac20452c91b4757526feb3e6 : SUCCESS in 1h 52m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3e2ec0b9458f4fdfa77c93a5ea48befd : TIMED_OUT in 3h 09m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03954e9a9a2e482ab20ef58578da7c16 : SUCCESS in 1h 31m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0fbcd457fc4fea984f8277fb90dbea : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":4},{"id":"9ab1e16875e64be165930aa35567f47e90811bac","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-11-29 16:25:17.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/4/20251129152012/NEC-FC : FAILURE in 10h 00m 26s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/4/20251129152012/NEC-ISCSI : FAILURE in 10h 05m 03s","accounts_in_message":[],"_revision_number":4},{"id":"54a2ff2e51cc5e5d9f67684b1065a3931a5ae2c3","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2025-12-01 09:40:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3123f4d88ef02de2e4e9037e8f10f4d76898bb28","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2025-12-03 15:18:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c9b1483df3a71ff72a13217e5e7793bff33e2316","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-05 15:19:08.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/4/20251205194713/NEC-V-ISCSI : FAILURE in 3h 50m 11s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/4/20251205194713/NEC-V-FC : FAILURE in 4h 31m 53s","accounts_in_message":[],"_revision_number":4},{"id":"3e43005159355642c88d0bb57b8ef09f21085b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-08 13:53:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0e04b1622e557dfb03d83d99580c01ebee0d6813","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-08 14:46:42.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":5},{"id":"b57e3904a57a47d46bfe4db1fbb0a07b72e8b6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-01-08 14:48:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a00b1d93ba724f3fa3635ff9e7e402e5 : FAILURE in 39m 20s","accounts_in_message":[],"_revision_number":5},{"id":"95437a4f3fcba447d31eeac7e83a30bbb360ba2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:30:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d6b5d6803a94bf3ad16ebfc04204799\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9d2aaed642f48899fb5501d859b0f10 : FAILURE in 9m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0dc82d2a601f497ab5a867ac4b241bec : FAILURE in 11m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1100a6dae8594cd8bdc15f6f6f014241 : FAILURE in 28m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16523cf6bf52bd689c288a407ea76f31d90e519d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-08 15:37:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/47db648a5a9c49688fbda7446423be69 : FAILURE in 1h 10m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6807391e816545169c40ade2437b92bb : RETRY_LIMIT in 15m 42s","accounts_in_message":[],"_revision_number":5},{"id":"f3f3da619e923a1dca44647a0844a6efb4f2b065","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-08 16:41:40.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7fdfaef0dea1fdf3d3fc3ffadf2b17d0496013eb","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-08 16:51:54.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_151/index.html : SUCCESS in 178m","accounts_in_message":[],"_revision_number":5},{"id":"a5d1663f01bd5debed46d0e607f4bc3c17217847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:15:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75e1159d585649d0825013b949661044\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f5e1d1b31b408182c0f486aeb5d79b : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bd2ceefaaf54bf88a079ca9f6ca9e50 : FAILURE in 8m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9511cadd3a9e431bbfaf2d573aff1a37 : FAILURE in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7308d0b429dc4fd8bd50a09831b6f467 : SUCCESS in 15m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/35153d06856b4102af8bb483723ac861 : SUCCESS in 29m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/136305fd3c664ca18dc4c53c2bd9ab59 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf7b6380024443e2a3512c00f2786f9c : SUCCESS in 1h 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b58efe7ddafe4600a126ffd618c4b225 : SUCCESS in 1h 03m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/be44d664c25c465ab6830303ef798362 : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e52127047084e3982a09191abdd1d1e : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/429690f7e4864edd9992f8d06f86c33f : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b9defea79f141219f38d7343e2211d8 : SUCCESS in 5m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aab51c7691094ae6a25e3080d5f2e788 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/41a8fd8a81924329b11d40ea54c7862b : FAILURE in 38m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e92fe2bb41d4996badd58e624908d68 : SUCCESS in 3m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c70b958cede84f5e9f1f8cb2d2039dee : SUCCESS in 1h 58m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cec4f9fa652d453b840daceceec1aab5 : FAILURE in 2h 51m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/effc7decc9df4b068d0439a054893287 : SUCCESS in 1h 49m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ac4ee59cd709478988668e280e1a9470 : POST_FAILURE in 1h 59m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/05a46edc18b341128fa2213b031ca544 : SUCCESS in 28m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db97ba93be2641a38e61dedd746d8e64 : SUCCESS in 40m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b07c0179296d4dedb3bb45984ab0889a : SUCCESS in 1h 32m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d530cfdb18a40e88ce34d63345cb074 : SUCCESS in 42m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e11bf320201a4883807cc9e966e329ae : SUCCESS in 1h 11m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e2fde6243d3a4f14bfc1a0b472d2d71d : SUCCESS in 1h 20m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/062d797218b044e79a06f2beeaabf6a7 : SUCCESS in 2h 08m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f37ebed57af44364a59dfc638bc5fa6d : SUCCESS in 39m 23s","accounts_in_message":[],"_revision_number":5},{"id":"d859c5f8aba5794a44410f703692ea86e01a8550","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-08 19:39:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 1h 19m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d5d5a855c90adf99a1271ef1ee473f6097128460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-12 05:04:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d6863a1c0abe797015b5109b536c0677b4475e24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-12 05:08:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fcfe575d804c4f4286892ec93f116fbf : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0fa13a12f5a84166825c2a62c5aafb0b : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"1d663a7aaebd939c37be32e3695d3211fae5f0a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 05:28:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e723eba1204a689c52db0d99730c2e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ecf2905f112b45cf874091fca819efc9 : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ac81e001b1a4cc68897760079dc3775 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9036bcca3e4424d821493962de6aa58 : SUCCESS in 22m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61bfee2fade78141c267918726f8fe1b38552c94","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-12 05:38:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/87258df3575a42869807c620e3308954 : FAILURE in 33m 43s","accounts_in_message":[],"_revision_number":6},{"id":"f7222d4e6c586c82316568b3c2ac8dbd926911f7","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-12 05:41:26.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6e0efbb910184278a17ee73e26d1e9ff : FAILURE in 36m 53s","accounts_in_message":[],"_revision_number":6},{"id":"60ba319c44ad70c87782b4b5e55768be847ff58e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-12 06:00:05.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":6},{"id":"7e1c57055625b18363c587680f9c80d14cd0f9c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-12 06:10:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b01401a23a3d843962cf5e8f0daef78b1a89fe21","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-12 06:14:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/663ab7a0ea2a4315a6cba0a6a08131e6 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7f0c486cb246478c9c364f203a1e474e : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"bfaee4fe0580c63ebc1c0858afec5dd00f8b2244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 06:26:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94c210061af0450ebe9b5722f905b94e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc3189258fc642769529698becb025b9 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0a1cd3d8b8043c4aaf869a0e925fd3c : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6ed850475994cbcb972b1327e46efb5 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"464ffad298ba09c83e4ed95f29eb2beb45c93730","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-12 07:08:54.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":7},{"id":"7db02214e5343fbde13f4fb53d8b1ee770a46c5d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-12 07:10:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/059e115b69b245e5accf522b915c580b : FAILURE in 30m 15s","accounts_in_message":[],"_revision_number":7},{"id":"91bfff6a50bf25dc0eb9ed3599d20210b30648cd","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-12 07:53:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/905c3fc0f28949df8da4968ec80db978 : FAILURE in 38m 39s","accounts_in_message":[],"_revision_number":7},{"id":"665602b6eb156feeb8e5639d3c34c6a6d19c8743","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-12 08:57:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 1h 19m 28s","accounts_in_message":[],"_revision_number":7},{"id":"966aa7a7478502643384218f2751ff652cd8118d","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-12 09:19:57.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_180/index.html : SUCCESS in 67m","accounts_in_message":[],"_revision_number":7},{"id":"dc066baf42a07cab0e68c5801f20ccf69c78b3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 09:26:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d123ce44c874f76adcad9745aa3296a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99de014ce174417a7a567ed41b549fd : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec4c469e3050461982a208ff7ae451ef : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3cd28137b4b54ba8979882e5b0485b86 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389cf0c3437e4a81ae0aa54c8f4d358d : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/344cb5e8df384c1d968fd529372abd05 : SUCCESS in 38m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed63f637b9b94edab0fb04d725dae033 : SUCCESS in 1h 03m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/699559276d7b41ca8f8ccde81bd48c2c : SUCCESS in 1h 39m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c4a6a3a21a42ca9fd17643e994fa36 : SUCCESS in 1h 03m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fc859836a714d898645c80890b5d12b : SUCCESS in 19m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/45c8731fd55a4bc78d4ff67d306894a7 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8998c9d1c244d3cb2f1ed9fb302ff35 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/217f0b22a4ff4cd8b53fd05f0907994f : SUCCESS in 4m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d58393030da34910ab0789a3675cf3fd : SUCCESS in 4m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/99be71eb875c417fa679b1ffdeeabb6b : FAILURE in 37m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9759c6cb997d4707b03ea66b8f0cabbc : SUCCESS in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/886e46b18c1a4f7f850b322d99d60a03 : FAILURE in 1h 13m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f4d26bd76dcf4460ae7cbcf76c599a54 : TIMED_OUT in 3h 04m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b81746d35130435088496c7ee93ee699 : SUCCESS in 1h 26m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/25d68428da4545799d24fa475307cc3b : SUCCESS in 1h 26m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/51d02d1937c84fe7bfeb5992a6bcac12 : SUCCESS in 44m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/395724d4156044ce999f07930f2a8002 : SUCCESS in 1h 17m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3dd570d99ad2467585d1c5d53f0a0f1c : SUCCESS in 1h 34m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ee8e3cfad0e64475bf2247d09bd4ac81 : SUCCESS in 43m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f11a893b7e754356a5f4f4ef021ce8f5 : SUCCESS in 2h 03m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9a5004cbc6d94570843740031a03fa50 : SUCCESS in 2h 05m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a45c8ae318ed488782833e92177c9651 : SUCCESS in 1h 31m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3711a2eea16649c886772c747175926e : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d66478a3edabd505068643083ddc7cb39cb2da40","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-14 06:21:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9e9b7034ba45a09919c1fc34359280a04b60ad21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 06:37:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e879cb1ca3474f9d4fe0ce91a5b969\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2e0a6ab477f461d972e64c33f2c52d1 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3ce156dc1f94428bb8e8c8bcacdad2c9 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2cd048486942404fa495929342a60e98 : SUCCESS in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"be8821ba1307cdd11c9a1c4cd6a581bfc0bd994c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:32:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b9fb236cd6e40e387e7099f8c6b74a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7dd3db08d4d4b5eb7f399a5b739b914 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fca747360c09411692176a4a73b69b24 : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a77d0e068449b484e002e54feda55e : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e99134d5edc43f2b70b16b8955f2731 : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/dc1a43b7c5294bdda3f494e39fd1c460 : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad2567c06db94c1eb7dc522fd0042baf : SUCCESS in 1h 01m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/551f04b1c7b74977b9472b8796c6bb0d : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ba7582a9d1e45e59af63cfcf2376259 : SUCCESS in 1h 02m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd81aaccb7014a62ae902a8669c32f53 : TIMED_OUT in 41m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e95db055bb6248779679fef9c6963605 : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c0bde4af2354df0b1e969812804559c : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87d40a6094624d5c9ff696024c2c8c70 : SUCCESS in 6m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4768bd80ae6a4a30848425f6e13cd033 : SUCCESS in 3m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c8863448bb12468f8716e0c95d8dc240 : FAILURE in 21m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/43b2778563434b799b9043a646b07645 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0dee0d1467c848b9946df983864cdb67 : SUCCESS in 1h 20m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3b533a8cd0346d7aa116eb6235f121a : FAILURE in 3h 04m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65d11deb801f4428a7a549ace7502637 : SUCCESS in 1h 50m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/386184e442a84639bb0165db968ebc17 : FAILURE in 1h 28m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0aaec23e2dbc410d88bf14f7c685cb22 : SUCCESS in 24m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/332bda65876540c39c12b7152e2dffbc : SUCCESS in 50m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c71e355255b745f8b452b34fba8fd145 : SUCCESS in 1h 35m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/60b119ad5ccf46649658c583d2a7f1e0 : SUCCESS in 24m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/67b8462f13fd42a5a7279bb7061b0be4 : SUCCESS in 1h 05m 37s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2f55aecc02c64f2298d8c2f3747bb881 : FAILURE in 1h 22m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86fcfe0fe7d14003864dfc68f61cb388 : SUCCESS in 1h 14m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea58af41c0a40b9aa8f93f266adc5ac : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":7},{"id":"4e92cdf542034a49b0447da3f674f5ef170ad062","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-14 09:39:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_01_2026_09_32_26\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"11104f9e7d94bf0e0a676017995ae6ad01936129","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-14 14:17:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-968831-7-fc : SUCCESS in 1h 15m 25s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-968831-7-iscsi : SUCCESS in 1h 15m 1s","accounts_in_message":[],"_revision_number":7},{"id":"496c70910fe5850c0f14aba67035896e018004e9","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-01-14 14:47:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c926ca19f4248fc471a7d5d5a4987f88f744a85b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-15 03:16:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e18f23707b24297014ff4dda9689f81c944563d6","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-15 03:49:33.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/48b9f101670846c5bcfeda8a2c5a6c6a : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":8},{"id":"2eedc32fc118e1958ef265f841bdf57bc1cc1c81","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-15 03:52:25.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"feef9e44bab09ea511227390baad5cdd18328c86","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-15 03:54:32.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/82c5c18bbee64cd8bc055107d96e2378 : FAILURE in 37m 40s","accounts_in_message":[],"_revision_number":8},{"id":"616049dc51e8395936edb926eb7f21f405486b8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 03:55:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2efbce5abf47f780edc8e45dd261a2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/514e744852014a3abb2d270e9b18f505 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/798e296859b24ec8b8433e6aef52d91d : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f52c64e57f5649109a804d5955292a53 : SUCCESS in 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dd584ce4c0264ea463bf6652fb7129a87c7a664d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-15 04:12:44.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":8},{"id":"64b3b984535964ee352d05420ed174018a540ae0","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-15 04:47:48.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_223/index.html : FAILURE in 91m","accounts_in_message":[],"_revision_number":8},{"id":"585e9565790332ca9ecec4a5aadfdf8aa04d9ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 05:32:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a632d9bea0a439a80ff010d4e7a9939\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d71dbb51a6c4eaf8c20908879c52fef : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b736c75642c419b830443d72aaa4ce5 : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79defae0de484328b284a13762deb484 : FAILURE in 15m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5624a89d3e4833ae501d2a6b3de149 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/439ac11473624f66ac02b833fe3ff50d : SUCCESS in 1h 07m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/739af495d1eb48d09b7e72024d0d70b9 : SUCCESS in 34m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45741de21dbf4aeb96c18f3de96b1504 : SUCCESS in 1h 37m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c1e5ca7f80540a3a39e2dd1d83f307f : SUCCESS in 1h 05m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e3da46eb5614f2a9383fa7592c967d0 : SUCCESS in 15m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/820978fefeaf40e79dd05bafe7597584 : SUCCESS in 25m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bceb113ca1944561bb305f333610d95e : SUCCESS in 3m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/407884de5853417e97dfcb4f24e06c60 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96fdd1e5c5334fff949edd7fcd5d9ae5 : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4563cdc4c1834071ab7bd79871ace8df : SUCCESS in 4m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ae15cf7a7fb4ace9a4537e58ab8c916 : FAILURE in 40m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ce521b9ddb44b96a8e4e4b8a6757baf : SUCCESS in 7m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a46b1f37e1764b82aa564acc2fc64227 : SUCCESS in 2h 01m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/065c4410bcb5473c877e51771a44d5eb : FAILURE in 1h 53m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5516cf2779ec4e33bf0311fe57b49e3d : SUCCESS in 1h 52m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b2fb298ce0a94fdbb64ee696c23ebe7e : SUCCESS in 2h 07m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4c141f4957c8416a90bfcb17d0d3ada4 : SUCCESS in 35m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d32e914d93434dde98552bca82c34089 : SUCCESS in 35m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/26af083c76704f16ab95f1d0d15349f1 : SUCCESS in 1h 22m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fdfe8d837f664befa50fdc6bda6694a7 : SUCCESS in 28m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7c0c69cbf3a942858ec32bcf72cba431 : SUCCESS in 1h 09m 38s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/faec0ce3ac3f42eead2c64b057c1381d : SUCCESS in 2h 00m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ddc4963a9f7435d885cb6eddececd9c : SUCCESS in 53m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a3cd813514046b1a07ca832b98042b9 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":8},{"id":"62d5a9a97cf007d23229667ada609365a5eed076","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-15 06:32:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 1h 17m 55s","accounts_in_message":[],"_revision_number":8},{"id":"b9acf39ffbf1049a3add597b9310d8ae91875e01","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-16 01:36:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bd4f3c013a184659b398af79714ccb02 : FAILURE in 1h 12m 05s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5adb33f5477a4ea9b7279996707264a0 : RETRY_LIMIT in 14m 57s","accounts_in_message":[],"_revision_number":8},{"id":"94b11f5d9d2fee9ebe8756adc1b9c6cceececc69","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-01-16 14:49:41.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c7a82d7c0c88f9e8a948c85767a165d41de80571","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-20 07:07:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a6d8b6fdfc2b975aaab6cfcd2ad06c275ce6d8ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 07:34:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eadfb0a1c844ed8bfa08992dceafc0b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3399e486e7194ae7b11d7d8118a2352e : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8e32b5e217243e38a1475ed4640d51e : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8756991ae6534983bf6d2bc4041df490 : RETRY_LIMIT in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9b45764ffce9c913a11f95d2624baa4b31bc7313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:10:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76bdcad7fc28483bb6e60566798e4a59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abcf852ffd9d426abeaa5b2070e01dce : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80f1c1fd679c4abeac8a611e480bec67 : SUCCESS in 7m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f935350f28e441b2acd87dca894767cf : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972689182a8543b8999cf2885beb8856 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/397037455b1443858e891ed028124766 : SUCCESS in 34m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f0e68fb1b9545608fbfe1f22fff7345 : SUCCESS in 34m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10e7181ac0544e0696045d48506b5cf8 : SUCCESS in 1h 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/235a0e099bd3473f99e765fba354ef69 : SUCCESS in 1h 02m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/981e603b1f8043acacfb3986799ad7ee : SUCCESS in 16m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/945d0c888cb44b039b56d312f2807105 : SUCCESS in 21m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f269907ba18843f2b0700c2f13d4e7db : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf5f8ad65c704ecf839040217812f9af : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b9395306d764961b8bcb1db125ea002 : SUCCESS in 5m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/94ce4f6f67eb48478c136b448b266775 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ace4e0dcad64928bdf31c4b6e9fe2d4 : FAILURE in 42m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf7964ad7dc34346aa01b0448e9ab3ff : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d03a2235dd174e01920a54a32871333c : SUCCESS in 1h 09m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/141363b415a442e6b071c5b0b0b68add : FAILURE in 1h 29m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d16329689a324328a06d2385dbe90fd0 : SUCCESS in 2h 02m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e0d4271b05dc416cb32b41d1553e8ce4 : SUCCESS in 1h 27m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a83fc244088746ebac9d4f7312da9ed6 : SUCCESS in 24m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a13d95a6f7f4c05b95200a3f0bb1dd5 : SUCCESS in 49m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f9031d85d214ae9a9a453b2d5ea6618 : SUCCESS in 1h 22m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b3ca2efa0c2143639bfd28ec03c960e6 : SUCCESS in 25m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/783e0e0feb7f4c2fb18c418342fdb7e8 : SUCCESS in 1h 09m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e93988f03d7a406bb0fe7b6e62ffadcb : SUCCESS in 1h 41m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5062c268a2284717a7fc19fda3dda0b7 : SUCCESS in 56m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aff93296c974ea9b120d021ecbcd9a9 : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":8},{"id":"3184c3cee2608d5138f1412427f36953929cbb64","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-20 09:22:26.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_01_2026_09_14_17\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"93cfb22c51eec590b7c8c3fb0463ace6977f48ed","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-20 16:14:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-968831-8-fc : SUCCESS in 1h 53m 45s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-968831-8-iscsi : SUCCESS in 1h 36m 36s","accounts_in_message":[],"_revision_number":8},{"id":"b3e7ae3ba8a0060c84abf11f7abb806ee9b66aec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-20 19:15:47.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5311628fbac94711977e92f0be2c1acb : FAILURE in 1h 02m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/bdd0b8c59fce480b86a2acf9e0b886fb : RETRY_LIMIT in 15m 15s","accounts_in_message":[],"_revision_number":8},{"id":"767fbcd01270de57ce4408e4d3a767e7759a3985","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-21 12:21:02.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b598687ded5816867a32768469379a6d0f523ca4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-22 15:36:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"566dd2e67a874327ab274ea27841e353fc06a66a","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-28 12:58:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"709dc09ac3bc7ee80d6711bf03dae6ae28999ddc","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-01-29 12:08:21.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5e136b5ee21ef9b1c9f3059e531d35bd69dedb7f","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-05 13:23:43.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9b358d775b47607e1dd6beb0fc770de289fd526e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-05 15:55:12.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"2243aa4db830c751243915e4433d9f4b07e6f8c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 15:56:01.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a83168306bae635f5b01275898415e3d8f124b2f","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-05 16:03:28.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4f3418c2f1fe8c3fdd395688f682ad1d25b8f8a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:16:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0148d74c394481cb3f3a07e8d3c0c0d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f314705a6a084ea1a027d07de0303e54 : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8ac51d1cc974e5cb5d87258713dbb52 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89ea52ca9a7a4973b19d5d5e2b37d246 : RETRY_LIMIT in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4f676999a691553fc9769c0c21745e09ca7b02aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 16:56:34.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":9},{"id":"50cdf2bd383342b73a5e028db7c5883f07229110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 18:30:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4532fb97e58545dfa6631401173240d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb34206aee4b4bdda6ac334567ca8b54 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f88a56de0be149ddbebcfe68e0129f6b : SUCCESS in 9m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0618d141fd4b458d807bff8f548eba5d : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c1c92df6c84c548508ca8dc13c57be : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/f255cd123e8248cfbb679cc2e4a6b15a : SUCCESS in 54m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b62e5bb393274268a21cd2731c734b54 : SUCCESS in 30m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5576369a6c3845f99ca6c452baf2558a : SUCCESS in 1h 29m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efdaa0945f3748c5bd28ab2935e1c52f : SUCCESS in 59m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a30e480f83414d1eae9032425abe2085 : SUCCESS in 16m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e47de8951c69434299b748d1abeabdde : SUCCESS in 16m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6fd6494b4bd847d8af19e7ccd6a001de : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4af6ce74367f46cbaab1667dafec88fb : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da62c722379b4d65863f9d5f3990053f : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2532af91c4c14498bf5cafe5ec49e61a : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a160db9ec0b24930a5a385e2e356ce56 : FAILURE in 39m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06ea9ac9cd024bfab190176718a8bdb0 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/282d4ef4ea3a4741a5b8b8240da2a55e : SUCCESS in 1h 51m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f85002d4b98a4727b88d6db7a81f1474 : FAILURE in 2h 30m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c1ab9bfab5140eca8bc7f7125b6818a : SUCCESS in 1h 56m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/92e2479ebf7d4a5c809af9a29e046733 : SUCCESS in 1h 49m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/419ff8186a0a4839b3a5ac052c47f5dd : SUCCESS in 31m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a73918d9f9bf4b1ab70ed6ec053943e1 : SUCCESS in 1h 14m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f716965020d14f9193011e86e6b48f55 : SUCCESS in 1h 28m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6954f400ea9e455295e54a1d79dff307 : SUCCESS in 41m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f0d9ceda4d1145d09b33319497acb422 : SUCCESS in 1h 06m 09s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4a9ba628e665465f9482480cda3dc8b3 : SUCCESS in 2h 00m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2502db4ac964fda9e1ab1e4c041d2b4 : SUCCESS in 1h 57m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79eeb711ae9542a0a85785224f50b4e1 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":9},{"id":"8dec74b31ae271f5b5052e4298efb095b63fbf9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 18:40:34.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_18_33_43\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"30485b837106a1a185246fc586279c5b2941a6a1","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 21:19:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d42c1f2c563c46b9a520d82f6adb588b : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":9},{"id":"e1313fb19ae241042ad7222e7f8d03bcac1222c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 22:44:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 39m 28s","accounts_in_message":[],"_revision_number":9},{"id":"ec25a23172d99a2abe227666c1eaa8f1137ebf46","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-06 06:52:55.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5a3db466b76463375fb60dc7f1813e5b55059312","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-10 04:18:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8e5425af35416b28f539595559506e13bcc68b54","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-10 04:26:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"64f43d14f1c4027ca279c07af1ea558b03e4d8a5","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-10 04:33:37.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ca36bf3fa9a2798db08500bf2838095bf7998338","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-10 04:36:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"85b4cf9a2189ef5b03a42d45fb596613e6914096","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-10 04:44:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"300e03b3268e8678b96859f436fa9860b01db8e4","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-12 16:32:23.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f9d33058a9d075d405ba1b64eae26297f7973665","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:30:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/15a809b3e29c4ddebba1b920f60e3103 : NODE_FAILURE Node request 200-0000623367 failed in 0s","accounts_in_message":[],"_revision_number":9},{"id":"089b87af5862c150e06cd9cdbbf3ebe7a4f0e117","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-02-13 15:52:23.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/9/20260213143616/NEC-ISCSI : FAILURE in 10h 12m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/968831/9/20260213143616/NEC-FC : FAILURE in 10h 16m 04s","accounts_in_message":[],"_revision_number":9},{"id":"ac0b16dc6415a226769aacafe540a78206abda14","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-02-16 14:23:33.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"92e96432ee3e2ede67a1718b3641dfd393062cd7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-16 14:44:33.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"94c666d8889aa566eaf02a0b5b91174b42d31056","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-17 04:48:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bdb652bc9c77edda0719364e8e1eab1457785533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 10:26:28.000000000","message":"Patch Set 10: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4b76996a88f1e359998b9a01538f6ca085839a72","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 10:32:26.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":10},{"id":"a86b7271f8f334358cf2449d5bb7b0537cab5507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 10:32:31.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"eabb548016e5d76f66e75bf7c522f8c4c95e1214","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 10:33:24.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"04138346e0c47bf0b347ae51b7461defc7a5758b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 10:38:25.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":11},{"id":"6d55065ee1b4c1a8df3b5f4670945c50abbd8b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 10:41:19.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"90a31e773f0aee54adbb5326efa4ed3d3de44689","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 10:47:16.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":12},{"id":"cb4aed1f16ad6385449bd8b1249fe2e8daf2ba36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 11:10:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"98dcb57e501d71de3ea07a8f013165cc811a075e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 11:16:24.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":13},{"id":"9a6f8cc1e8e995de5ce74953c4c3d93cb02d3a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 11:28:25.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"d2ceaeed6bf0351037c8cc823db1190d2a35716e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 11:34:20.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":14},{"id":"33819c395d8b0f50f8daa4dab9c2e48b7402cc9f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-17 12:11:21.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 38m 29s","accounts_in_message":[],"_revision_number":14},{"id":"0ec873181c31710ff8c3ddfbbd927a66849f8766","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-17 12:13:55.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5ad854ed96ab4f3ca36112657921ded2 : FAILURE in 42m 08s","accounts_in_message":[],"_revision_number":14},{"id":"fd79bbe4a4941b4c75b6ff93b513e5206da72227","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-17 12:22:08.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/be873ff769274b489c219cd112ce7170 : FAILURE in 38m 36s","accounts_in_message":[],"_revision_number":14},{"id":"0a0671ba427de470033f57201b9cec074c1d587a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 13:33:51.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09bd496156804f8eb7395dd713c2a6d2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a2bed478ac84364b6359745ac758d3e : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22b2db3d03d34890897e25ddff9a28b6 : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43ba640c379e4b4da9dab800855c520a : RETRY_LIMIT in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"70445908e78a074c777e576a7df201abadb96e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 13:46:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df15415b61824b5e9aacfeba1279e8c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5835206483f14d7c977e836302f6a139 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4622f40b94c4da4a06e5110b310f7ae : SUCCESS in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8789558de4904973b434143f6023f5b8 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5742b84052e64971a8c63f57ad330313 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/a951788f540e481084f1cd48ca63408e : SUCCESS in 55m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4b0a0a44fe540b6ace488530f6f28dc : SUCCESS in 1h 01m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0581a1d771f48f9acb1c058124f7e39 : SUCCESS in 1h 23m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d59a01ea5c47475ea5ef7b9b053c90c8 : SUCCESS in 37m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/105eff0575114587835d30e23a15c56b : SUCCESS in 17m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1a5d7b1403e84ff9b3ad7d791c06e350 : SUCCESS in 24m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ca3da78cc2a4c63b16fd3b0937e65ac : SUCCESS in 6m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf691ff3de9e44d898646087f95ff848 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b971f82ec2544471bb5044a4cb99c847 : SUCCESS in 7m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/57db855c30af47118a720c4cd30327bf : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a7b5a74949d49bcbd7e2038970f4f32 : FAILURE in 24m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e681db18513a450797a6286732169450 : SUCCESS in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f71d215a831944e398e69c292f812836 : SUCCESS in 1h 41m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6302ad987c7845b4bc5911a25486577f : SUCCESS in 2h 08m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a559428bb91402e8fd2a2c7609b5f31 : SUCCESS in 2h 04m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b49b0593f2ab48d8b218b279dd2b7556 : SUCCESS in 2h 07m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b3e42b434e7a4d8fb47cefd44995d183 : SUCCESS in 41m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de48517fbb614ef6a50f2b8ae76d1b08 : SUCCESS in 1h 13m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f3d1ad25c5a4b55a849369732ce0a8a : SUCCESS in 1h 35m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/322c6df26943415f8d6390d3252067b3 : SUCCESS in 30m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6f0ae747c146428baf02800802031702 : FAILURE in 1h 45m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0230e19ef0c642f6a36ec0b5e1508841 : SUCCESS in 1h 59m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/803706706d9747668d73e256ab5518bd : SUCCESS in 1h 32m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9877276c31e4a05ac432c797cceb7bb : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":14},{"id":"be9b7e9ef3bfb750315b5bfd400b00a8c13c0c37","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-17 16:18:22.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_695/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":10},{"id":"2c703944df7e704b103ff9350ff771bcf5d93c58","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-17 18:01:06.000000000","message":"Patch Set 11:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_696/index.html : FAILURE in 102m","accounts_in_message":[],"_revision_number":11},{"id":"8e750e8533ecadb76b5f98ac0ff4de3085a74b40","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-17 19:36:31.000000000","message":"Patch Set 12:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_697/index.html : SUCCESS in 95m","accounts_in_message":[],"_revision_number":12},{"id":"31eac02d9239b0f1414a0a67057d02f3866c7246","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-17 20:21:24.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d31e3843cd1406c8be980d18daf94e5 : FAILURE in 14m 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cc64d415e2894900bb152c568d4e7a98 : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":14},{"id":"daf3da77f5a2349795ce983d3b7acf6fabbfdaeb","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-17 21:05:36.000000000","message":"Patch Set 13:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_698/index.html : SUCCESS in 88m","accounts_in_message":[],"_revision_number":13},{"id":"bab11803eb3afa2ce44d4f2cf085eaba766c0a2b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-17 22:37:58.000000000","message":"Patch Set 14:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_699/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":14},{"id":"fa72cb937ef858c03e7a688309f6d510657e81dc","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-19 05:56:08.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3dd98f1ccd246acc45d9eb24ab1392e74ad19f15","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-02-25 07:30:55.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"29a95e535c7e02dd33df47add0fa3c66d6ace62d","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-03 05:20:22.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d95388014f73a565a0c4d31c5125cf1c4d560e87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-06 14:30:21.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"48639f24eab18b71d193929c146bbc1be013ef05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-06 14:35:55.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ae8e72a643a34a96abcc955de562c0b0 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4c6e5b61e88244bdbd1a0590776d7f8c : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":15},{"id":"a321809942f78b95975056ca5b331b9a6591c861","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-06 14:43:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/bcb2e659fa3b403c91252f791a4474af : FAILURE in 12m 53s","accounts_in_message":[],"_revision_number":15},{"id":"82b01c4c95e1769eb1991cb34899426496963503","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-06 15:02:53.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hpe-fc-driver build/781ec6a8459f49f583c2dd1bb274d0fd : FAILURE in 32m 06s","accounts_in_message":[],"_revision_number":15},{"id":"93bb300405d96a2b23306bbd3535fb3d9fa36d3d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-06 15:04:14.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":15},{"id":"08b14f371559ef75d31b4f205db5fb2d2aa2c053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 15:10:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6ca5d6128c47f381a20f026d5496ae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df8375ba05d94fd0a59259b433d1812d : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05957f0278bf4512903501fcebc5fb02 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ee688b2ea1174563942013ba790a42e6 : SUCCESS in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"92362da2e76feb536b721176a2cc759e61b4bd94","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-06 15:16:10.000000000","message":"Patch Set 15:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/968831 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"fb08ba08060f3ba80d7cacd1c565e21524dacee8","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-06 15:17:04.000000000","message":"Patch Set 15:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/968831 : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"8bed482f01dd3143ff914081f3dd693bae98f932","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-06 15:19:38.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/968831 : FAILURE in 40m 16s","accounts_in_message":[],"_revision_number":15},{"id":"b08dc670c3d4f7d70da6e2ea6a1a58ec2c972559","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-06 15:19:51.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/323a2007a6c542e796464021c3e48015 : FAILURE in 39m 41s","accounts_in_message":[],"_revision_number":15},{"id":"d3f11ae71066aa16dbc6b8381ffb4ada0dfccfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 21:16:40.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ec9f1830762408cb56be86ede8fb82b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f418ea5ffcf4c0dbff1e62c149ccc77 : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c72d506562d04896a104833cc38216e6 : SUCCESS in 10m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e03d2dc517243ae8270e4e02f6bd140 : SUCCESS in 9m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b1eabf0864b477cac0f2967162aec8f : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc31ba2579b463289860348647a05c9 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/851003f2cec04bbbb9a755755d97cf7c : SUCCESS in 1h 04m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/578a99d22bb949e6abd94f035f337997 : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30db482d37f74f2a8ece699fc6ac042f : SUCCESS in 1h 40m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6efe8d6e522342b4a7c9d0d131f6e510 : SUCCESS in 39m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a99e22687f4867b4a6443712ed4a74 : SUCCESS in 15m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b2da937d694e4a6088a9acfe75ae08ac : FAILURE in 6m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b69abf9b62654585bfa85b8ebcc45bf0 : SUCCESS in 4m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/99f589afed5e438a9a75307e01b51b3c : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ab9e64a1a4d5448babffdc11cdbed9b0 : SUCCESS in 5m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ab742eed56314121abbfe4fffd055194 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd4768ebfc644337b3e75bf01e2154f6 : FAILURE in 47m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3861560d8a84b0094cf4b58c2f1fd24 : FAILURE in 4m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f3c69221a21c42f2a837394b851ffdca : FAILURE in 1h 12m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/504b3f4515fb480c80fe488253a892f2 : FAILURE in 2h 26m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63ae553ac46f4d37b97efe1260f19e5e : SUCCESS in 1h 39m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6970addadf8640eb96ac14f36c6c6fe7 : SUCCESS in 1h 58m 52s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c8fa1957180f4d7b9a9e099cb0c0f247 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f7728868891c45f1bbc8112a1765c0b7 : SUCCESS in 1h 11m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7342e4806da94985b5599bef9d8d0aa8 : FAILURE in 2h 03m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0920b382250248d1aa97bfa63b494d07 : SUCCESS in 35m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/39a89080a23d4a5b80da7c0e86d8d652 : SUCCESS in 1h 44m 15s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1713d4b9cf1d41c79459a10d15ee2548 : TIMED_OUT in 3h 36m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fef8e857262949368dc21de62e52e090 : SUCCESS in 2h 05m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f95f9acbefc49dc8b69a745e51aebe3 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":15},{"id":"f759b5b1ab16c117b78350311f1225776c42052f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-07 05:38:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3372/.","accounts_in_message":[],"_revision_number":15},{"id":"4f7085b9aeb7c90b47556f781de05b3292809bc0","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-07 15:20:59.000000000","message":"Patch Set 15:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_996/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":15},{"id":"828e7a87038401a41336a12ed8ac3b2bc69cf9a2","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-04-24 14:27:01.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dd9f9cd41f4771dbd4090c23ddffff03f91fe50d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-30 12:41:42.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"29a8f6eb934d8039acd1e35c9cb27ec75e905881","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-30 12:46:23.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c012c6c0c0ac90f6ffbdd35e85f7e8267f9ebaea","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-30 12:47:22.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1b9cead6b05751bac635fda4b12d86d72003cadf","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-30 13:36:02.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hpe-fc-driver build/3d1aff40803d4e648f16800a17bd0ae8 : FAILURE in 29m 45s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/968831/etc/tempest_log.log","accounts_in_message":[],"_revision_number":16},{"id":"2b84e1483c132fad018f277571aaac97f9a21307","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-30 13:43:21.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I81e5198089b3278e84e45a37c7734c45c3104e4a .","accounts_in_message":[],"_revision_number":16},{"id":"0548d447c96e4b0ac07ba7aa09de7bd8f1a8a30f","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-30 14:00:15.000000000","message":"Patch Set 16:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/968831 : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"59a0ade34418607e3df6964a8e8b1641eb2181c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:01:43.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e54f61a6004e4d90a1a3518205881b48\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f182c3c31a804d78b19ff8acfc66c673 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc1b14e74255458b82a3cc35e63ee5bd : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1fec54faf84c4a0db261f6b46c69ad19 : SUCCESS in 23m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"de48715c24e6a2ddf047ba0672b72cebd6f47d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:48:22.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d27c965804949079da4b1d375d3ed19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3218c0f32db94e10974927dadd01619f : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42926169e79b477ca58cd02cd6e68fdf : SUCCESS in 9m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8776b52cdc5b463db4452c4027f5626c : SUCCESS in 10m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e50ef99c1a9f4038bb281cd79305f75e : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c8ae7b26a214a6ab101b1aa54d48597 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/aa583be09de84a49a48dc947525c451a : SUCCESS in 1h 07m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3757296140b143a2a8ed47b01fe4e22e : SUCCESS in 1h 08m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f894bcde149e455b89358896d6eaba7f : SUCCESS in 1h 24m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8276c743c484b9e98f063f33aa8a55f : SUCCESS in 1h 03m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/183e42bac71e48baa4a459dacc84a8ff : SUCCESS in 21m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17cbf36085f04992b65ecd3229d76888 : SUCCESS in 23m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6659a4fac5a148f89143561721610d60 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50475c517a434c8893a3d6f31f2bf8d5 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44a3fa7e4da849be97fa5a096682885b : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b4154fc5c6041ada15a0033f0a2a5d0 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/95f58c90ece14f6293869dc3ead2f9d5 : SUCCESS in 43m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/559f2362ff2d4184b4e58f8e9b7bf259 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/509c73e6a783428f9a59008313284cf8 : SUCCESS in 1h 40m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3443f99a04af4c7f99c8c074a5153579 : FAILURE in 1h 49m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d59ca71d4f1f437780bf0eb9740dae64 : SUCCESS in 2h 00m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/acd431f3a55d4894a7a34b009fe0da60 : SUCCESS in 1h 58m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/91e3527b16604c45b8b63a3ddab6bf3f : SUCCESS in 20m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/43e6daa6953443fc912a0cfcb9b9043b : SUCCESS in 1h 16m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e6088da9a734d93bde8831e72d70389 : SUCCESS in 1h 25m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/082bd869980f4a5a9c39ffe73cf6a06e : SUCCESS in 46m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eb039d0d76774357906482599db32001 : SUCCESS in 1h 08m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66230e1c59f24a39b18c6cd3f50da0a3 : SUCCESS in 1h 44m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/741a9848661645308a69962da61585f6 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":16},{"id":"aa8f383d9f8d67cb4427c5d9824394513305953a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 14:54:48.000000000","message":"Patch Set 16:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3845/.","accounts_in_message":[],"_revision_number":16},{"id":"8332e085fc5c4b4160bc746591b799c78a97f4ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-30 14:57:49.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_04_2026_14_51_30\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"db5d8ebf334bf2b0db436d1b14d8b51e397147d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 15:12:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4021/.","accounts_in_message":[],"_revision_number":16},{"id":"35a01978bc01e71b72b02c0a265a337abd7136ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 04:54:59.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 968831,16","accounts_in_message":[],"_revision_number":16},{"id":"81b2c570620abb66849b14fdf99e0d3ccbbb53be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:00.000000000","message":"Patch Set 16:\n\nBuild Failed","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"19bdf7102e367ec7bdf32f8ce10f9e77b6a7fe81","revisions":{"7e181ab6938791844d4106e79334ae241119a62d":{"kind":"REWORK","_number":1,"created":"2025-11-28 17:47:22.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/1"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:46:49.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_reports_physical_capacity option for accurate capacity reporting.\n\nIntroduce netapp_driver_reports_physical_capacity boolean config option\nto enable reporting of physical used space (flexvol space.physical_used)\nas provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nNote: Enabling this feature increases API calls to storage and may\nimpact volume report performance.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e181ab6938791844d4106e79334ae241119a62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e181ab6938791844d4106e79334ae241119a62d"}]},"branch":"refs/heads/master"},"e976249c7fd19e4052506fb4ee279e204bd8d1f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-28 17:48:35.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/2"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:48:13.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_reports_physical_capacity option for accurate capacity reporting.\n\nIntroduce netapp_driver_reports_physical_capacity boolean config option\nto enable reporting of physical used space (flexvol space.physical_used)\nas provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e976249c7fd19e4052506fb4ee279e204bd8d1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e976249c7fd19e4052506fb4ee279e204bd8d1f1"}]},"branch":"refs/heads/master"},"c875c72e155679b182403ad2d670992158b3face":{"kind":"REWORK","_number":3,"created":"2025-11-28 18:03:38.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/3"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 18:03:23.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_reports_physical_capacity option for accurate capacity reporting.\n\nIntroduce netapp_driver_reports_physical_capacity boolean config option\nto enable reporting of physical used space (flexvol space.physical_used)\nas provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c875c72e155679b182403ad2d670992158b3face"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c875c72e155679b182403ad2d670992158b3face"}]},"branch":"refs/heads/master"},"6c85abf9ad101d7cebbaaad2f125893635224a78":{"kind":"REWORK","_number":4,"created":"2025-11-28 18:07:26.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/4"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 18:07:13.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_reports_physical_capacity option for accurate capacity reporting.\n\nIntroduce netapp_driver_reports_physical_capacity boolean config option\nto enable reporting of physical used space (flexvol space.physical_used)\nas provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c85abf9ad101d7cebbaaad2f125893635224a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c85abf9ad101d7cebbaaad2f125893635224a78"}]},"branch":"refs/heads/master"},"be414a7b254e9fb6537cbcac454ffa550d884eb2":{"kind":"REWORK","_number":5,"created":"2026-01-08 13:53:24.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/5"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-01-08 13:45:24.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_discount_clones_capacity option for clone discounted capacity reporting.\n\nIntroduce netapp_driver_discount_clones_capacity boolean config option\nto enable reporting volume(i.e.) LUNs capacity consumed at ONTAP. This\nachived by discarding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be414a7b254e9fb6537cbcac454ffa550d884eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be414a7b254e9fb6537cbcac454ffa550d884eb2"}]},"branch":"refs/heads/master"},"16ea6e70a19014b1ce606f90292eeca1c58ddab0":{"kind":"REWORK","_number":6,"created":"2026-01-12 05:04:07.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/6"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-01-12 05:03:49.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_discount_clones_capacity option for clone discounted capacity reporting.\n\nIntroduce netapp_driver_discount_clones_capacity boolean config option\nto enable reporting volume(i.e.) LUNs capacity consumed at ONTAP. This\nachived by discarding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16ea6e70a19014b1ce606f90292eeca1c58ddab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16ea6e70a19014b1ce606f90292eeca1c58ddab0"}]},"branch":"refs/heads/master"},"302320b76513a249a079fcfb15263ec0f7cd421f":{"kind":"REWORK","_number":7,"created":"2026-01-12 06:10:39.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/7"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-01-12 06:10:15.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_discount_clones_capacity option for clone discounted capacity reporting.\n\nIntroduce netapp_driver_discount_clones_capacity boolean config option\nto enable reporting volume(i.e.) LUNs capacity consumed at ONTAP. This\nachived by discarding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/302320b76513a249a079fcfb15263ec0f7cd421f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/302320b76513a249a079fcfb15263ec0f7cd421f"}]},"branch":"refs/heads/master"},"01d1018c960cfc5c799d8f3356d091bf0c1c770f":{"kind":"REWORK","_number":8,"created":"2026-01-15 03:16:28.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/8"}}},"commit":{"parents":[{"commit":"e23e859dfd570ff1bd784ec62fc0258ca5681dee","subject":"Merge \"Show volume attachment host information for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e23e859dfd570ff1bd784ec62fc0258ca5681dee"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-01-15 03:16:03.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_discount_clones_capacity option for clone discounted capacity reporting.\n\nIntroduce netapp_driver_discount_clones_capacity boolean config option\nto enable reporting volume(i.e.) LUNs capacity consumed at ONTAP. This\nachived by discarding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option takes precedence over\nnetapp_driver_reports_provisioned_capacity.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp. Enabling this option queries ONTAP flexvol\nphysical used space instead of calculating the size of each LUN/file.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01d1018c960cfc5c799d8f3356d091bf0c1c770f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01d1018c960cfc5c799d8f3356d091bf0c1c770f"}]},"branch":"refs/heads/master"},"2623c2c046120eaf86ed04a4831b61219a21236f":{"kind":"REWORK","_number":9,"created":"2026-02-05 15:55:12.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/9"}}},"commit":{"parents":[{"commit":"27af2c644dd9982162549a5c71c5c8ab458c4543","subject":"Merge \"Default disk_geometry option to unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27af2c644dd9982162549a5c71c5c8ab458c4543"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-02-05 15:54:42.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nThis addresses scenarios like Veritas NetBackup workflows where volumes\nare created from clone-of-clone operations that don\u0027t consume additional\nphysical storage at NetApp.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2623c2c046120eaf86ed04a4831b61219a21236f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2623c2c046120eaf86ed04a4831b61219a21236f"}]},"branch":"refs/heads/master"},"434f8af55fd565a6cf0d848b00b1c344224660f8":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-02-17 10:26:28.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/10"}}},"commit":{"parents":[{"commit":"27af2c644dd9982162549a5c71c5c8ab458c4543","subject":"Merge \"Default disk_geometry option to unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27af2c644dd9982162549a5c71c5c8ab458c4543"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanand borra","email":"jayaanand.borra@netapp.com","date":"2026-02-17 10:26:28.000000000","tz":0},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot) for backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that consume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks. However, Cinder counts each clone at full logical size in provisioned_capacity_gb. Example: 1 source volume (10TB) + 2 NetBackup clone snapshots \u003d 30TB reported provisioned capacity, while actual physical consumption remains ~10TB. This causes the CapacityFilter to reject new volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude ONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb, reporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/434f8af55fd565a6cf0d848b00b1c344224660f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/434f8af55fd565a6cf0d848b00b1c344224660f8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1b2c53b6c3f9b498d8ddd6c72f7fae6e0bdf485c":{"kind":"REWORK","_number":11,"created":"2026-02-17 10:32:31.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/11"}}},"commit":{"parents":[{"commit":"27af2c644dd9982162549a5c71c5c8ab458c4543","subject":"Merge \"Default disk_geometry option to unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27af2c644dd9982162549a5c71c5c8ab458c4543"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanand borra","email":"jayaanand.borra@netapp.com","date":"2026-02-17 10:32:31.000000000","tz":0},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot) for backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that consume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks. However, Cinder counts each clone at full logical size in provisioned_capacity_gb. Example: 1 source volume (10TB) + 2 NetBackup clone snapshots \u003d 30TB reported provisioned capacity, while actual physical consumption remains ~10TB. This causes the CapacityFilter to reject new volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude ONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb, reporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b2c53b6c3f9b498d8ddd6c72f7fae6e0bdf485c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b2c53b6c3f9b498d8ddd6c72f7fae6e0bdf485c"}]},"branch":"refs/heads/master"},"72fa8bd026baf391d49d44c7cf03ed6456328e16":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-02-17 10:41:19.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/12"}}},"commit":{"parents":[{"commit":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","subject":"Remove Windows drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanand borra","email":"jayaanand.borra@netapp.com","date":"2026-02-17 10:41:19.000000000","tz":0},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot) for backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that consume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks. However, Cinder counts each clone at full logical size in provisioned_capacity_gb. Example: 1 source volume (10TB) + 2 NetBackup clone snapshots \u003d 30TB reported provisioned capacity, while actual physical consumption remains ~10TB. This causes the CapacityFilter to reject new volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude ONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb, reporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72fa8bd026baf391d49d44c7cf03ed6456328e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72fa8bd026baf391d49d44c7cf03ed6456328e16"}]},"branch":"refs/heads/master","description":"Rebase"},"d337faa857f9d0fc3ad29e3eb4b7bcb6db840117":{"kind":"REWORK","_number":13,"created":"2026-02-17 11:10:31.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/13"}}},"commit":{"parents":[{"commit":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","subject":"Remove Windows drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-02-17 11:08:09.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot)\nfor backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that\nconsume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks.\nHowever, Cinder counts each clone at full logical size in provisioned_capacity_gb.\nExample: 1 source volume (10TB) + 2 NetBackup clone snapshots \u003d 30TB reported provisioned capacity,\nwhile actual physical consumption remains ~10TB. This causes the CapacityFilter to reject\nnew volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude\nONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb,\nreporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d337faa857f9d0fc3ad29e3eb4b7bcb6db840117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d337faa857f9d0fc3ad29e3eb4b7bcb6db840117"}]},"branch":"refs/heads/master"},"4d288e1151ca9bf85f5520879bf3a37801e62716":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-02-17 11:28:25.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/14"}}},"commit":{"parents":[{"commit":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","subject":"Remove Windows drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-02-17 11:25:48.000000000","tz":-300},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot)\nfor backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that\nconsume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks.\nHowever, Cinder counts each clone at full logical size in provisioned_capacity_gb.\nExample: 1 source volume (10TB) + 2 NetBackup clone-of-clone(i.e cinder volume and ONTAP LUN)\u003d30TB\nreported provisioned capacity, while actual physical consumption remains ~10TB.\nAdditional capacity may be consumed as user writes data. This causes the CapacityFilter to reject\nnew volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude\nONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb,\nreporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d288e1151ca9bf85f5520879bf3a37801e62716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d288e1151ca9bf85f5520879bf3a37801e62716"}]},"branch":"refs/heads/master"},"7b4593a0341370c877a4a3b5082b77a492076799":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-03-06 14:30:21.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/15"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2025-11-28 17:30:46.000000000","tz":-300},"committer":{"name":"jayaanand borra","email":"jayaanand.borra@netapp.com","date":"2026-03-06 14:30:21.000000000","tz":0},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded capacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from snapshots)\nfrom the provisioned capacity calculation, reporting only non-clone storage\nconsumption as provisioned_capacity_gb. This option requires\nnetapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack volume create --snapshot)\nfor backup operations. On NetApp ONTAP, these are clone-of-clone (COW) volumes that\nconsume zero additional physical storage — they are metadata pointers to the parent\u0027s data blocks.\nHowever, Cinder counts each clone at full logical size in provisioned_capacity_gb.\nExample: 1 source volume (10TB) + 2 NetBackup clone-of-clone(i.e cinder volume and ONTAP LUN)\u003d30TB\nreported provisioned capacity, while actual physical consumption remains ~10TB.\nAdditional capacity may be consumed as user writes data. This causes the CapacityFilter to reject\nnew volume creation requests because it believes the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp driver to exclude\nONTAP clones (identified via the is-clone attribute) from provisioned_capacity_gb,\nreporting only non-clone storage consumption. This is an opt-in driver configuration.\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b4593a0341370c877a4a3b5082b77a492076799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b4593a0341370c877a4a3b5082b77a492076799"}]},"branch":"refs/heads/master","description":"Rebase"},"19bdf7102e367ec7bdf32f8ce10f9e77b6a7fe81":{"kind":"REWORK","_number":16,"created":"2026-04-30 12:41:42.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/31/968831/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/968831/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/968831/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/968831/16"}}},"commit":{"parents":[{"commit":"65bba6830d578d7fe3138dfa65228dfda19eb49e","subject":"Merge \"Add support for AutomatedFailOver policy with consistent replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65bba6830d578d7fe3138dfa65228dfda19eb49e"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-04-30 12:39:03.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-04-30 12:39:03.000000000","tz":-240},"subject":"NetApp: Report used capacity consumed at FlexVol to Cinder pool","message":"NetApp: Report used capacity consumed at FlexVol to Cinder pool\n\nAdd netapp_driver_exclude_clones_capacity option for clone-excluded\ncapacity reporting.\n\nIntroduce netapp_driver_exclude_clones_capacity boolean config option\nto enable reporting volume (i.e., LUNs) capacity consumed at ONTAP. This\nis achieved by excluding all clones (including volumes created from\nsnapshots) from the provisioned capacity calculation, reporting only\nnon-clone storage consumption as provisioned_capacity_gb. This option\nrequires netapp_driver_reports_provisioned_capacity to also be enabled.\n\nWhen netapp_driver_exclude_clones_capacity is enabled without\nnetapp_driver_reports_provisioned_capacity, a warning is logged to\nalert the user that both options must be enabled for clone exclusion\nto take effect.\n\nVeritas NetBackup creates Cinder volumes from snapshots (openstack\nvolume create --snapshot) for backup operations. On NetApp ONTAP, these\nare clone-of-clone (COW) volumes that consume zero additional physical\nstorage - they are metadata pointers to the parent\u0027s data blocks.\nHowever, Cinder counts each clone at full logical size in\nprovisioned_capacity_gb. Example: 1 source volume (10TB) + 2 NetBackup\nclone-of-clone (i.e. cinder volume and ONTAP LUN) \u003d 30TB reported\nprovisioned capacity, while actual physical consumption remains ~10TB.\nAdditional capacity may be consumed as users write data. This causes\nthe CapacityFilter to reject new volume creation requests because it\nbelieves the backend is overprovisioned.\n\nThe netapp_driver_exclude_clones_capacity option enables the NetApp\ndriver to exclude ONTAP clones (identified via the is-clone attribute)\nfrom provisioned_capacity_gb, reporting only non-clone storage\nconsumption. This is an opt-in driver configuration.\n\nREST/ZAPI hybrid for is-clone lookup\n------------------------------------\nThe ONTAP REST API does not currently expose the is-clone attribute on\n/storage/luns. Rather than dropping REST support for LUN size listing\nentirely, the REST client retains its native /storage/luns path for the\ndefault (non-exclusion) case and transparently falls back to the\ninternal ZAPI client only when exclude_clones\u003dTrue is requested. This\nfollows the existing \"REST method not supported\" fallback pattern in\nclient_cmode_rest.RestClient (which already instantiates an internal\nzapi_client for exactly this purpose) and avoids regressing REST-only\ndeployments that do not opt in to clone exclusion. The fallback will be\nremoved once ONTAP REST exposes is-clone for LUNs.\n\nCaveats documented in the release note\n--------------------------------------\nONTAP FlexClones consume virtually zero physical space at creation but\ngrow over time as the active volume\u0027s referenced blocks are overwritten\n(copy-on-write). When netapp_driver_exclude_clones_capacity is enabled,\nthe reported provisioned_capacity_gb therefore reflects clone footprint\nat creation time and may under-report true provisioned capacity as\nclones are written to. The release note now calls this out explicitly\nand recommends pairing the option with conservative\nmax_over_subscription_ratio values for clone-heavy workflows\n(e.g. Veritas NetBackup).\n\nChange-Id: I81e5198089b3278e84e45a37c7734c45c3104e4a\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19bdf7102e367ec7bdf32f8ce10f9e77b6a7fe81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19bdf7102e367ec7bdf32f8ce10f9e77b6a7fe81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
