)]}'
{"id":"openstack%2Fcinder~969493","triplet_id":"openstack%2Fcinder~master~Idbe38bd27fa213069ee26305e5f519b9f5dfcb38","project":"openstack/cinder","branch":"master","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2025-12-03 18:30:08.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}}},"removed_from_attention_set":{"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2025-12-03 18:25:35.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Idbe38bd27fa213069ee26305e5f519b9f5dfcb38","subject":"feat: use image multipart download","status":"NEW","created":"2025-12-03 16:45:09.000000000","updated":"2025-12-13 18:36:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"b9e2115a716d366d7499b32d0c29cc45710e5051","_number":969493,"virtual_id_number":969493,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-05 13:19:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-12-03 16:45:31.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-12-03 16:49:09.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-03 17:43:10.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 17:48:06.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 18:00:48.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-12-03 18:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 18:23:03.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-12-03 18:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 19:48:36.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-12-13 00:55:11.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-12-13 18:36:47.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cd2a9c84c8a1f4335104909b31633c4768853678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-03 16:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d86a7f60789bed9911c9c0bb357b7de39a08238e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 16:45:31.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2621/.","accounts_in_message":[],"_revision_number":1},{"id":"d35a433f0a2cec51f200f772331aca9d2ed3505f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 16:45:36.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2769/.","accounts_in_message":[],"_revision_number":1},{"id":"f0f72492264bc5bfcfa5437cbeb3cf01a06d1a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-03 16:49:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bafc4fbb3cbca93694c52e10bffb9eac14dac7f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-03 16:49:09.000000000","message":"Patch Set 1:\n\nBuild Aborted \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/118/ : ABORTED (skipped)","accounts_in_message":[],"_revision_number":1},{"id":"e162c87a56ff04f2710c51b7d59e094de32d1680","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 16:49:20.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2770/.","accounts_in_message":[],"_revision_number":2},{"id":"3dd3b5002c5955d31ee73d42bab3ab0494e4090b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 16:49:20.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2622/.","accounts_in_message":[],"_revision_number":2},{"id":"d2bff21405d7ca209b5465b9ebf98b7ca094f76e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-03 17:00:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/119/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"83288d3a93716ecde1cdf0687c6e4db32f4dcaa4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-03 17:43:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8133335cdb1c48019e59bcd46280d1ce : SUCCESS in 53m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d765d9568f394c209d6b2c545930e4a9 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":2},{"id":"72541cf344e1de06e29994ad88fbefb259256e92","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-03 17:48:06.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idbe38bd27fa213069ee26305e5f519b9f5dfcb38 .","accounts_in_message":[],"_revision_number":2},{"id":"cc4c4e3a4d2c4583ccddc332b1bf485de1ca4bc1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-03 17:50:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idbe38bd27fa213069ee26305e5f519b9f5dfcb38 .","accounts_in_message":[],"_revision_number":1},{"id":"df1f460ecf17c632e89bec83902cf26c0e62ade5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-03 18:00:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/969493 : FAILURE in 1h 06m 48s","accounts_in_message":[],"_revision_number":2},{"id":"95d856d4a561d245b9126b1293b5e7212e30855a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:17:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fbe7603e4b4698be4bd6d97b4d8993\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c205e7e081714b66b31c03035e24527c : SUCCESS in 31m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/df3630d1dff04eb2826bbe71cc700594 : SUCCESS in 37m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1120c05fbb044ef9833f521f57de5721 : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"640a2a2fa54f6a2f6ac544082a5e46e29b3b3d88","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-12-03 18:23:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a4b7d22211acb488251eda5ecdd25d2c505c144","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-03 18:25:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"ae6479bb6f97b02a1f3e977c660340c8f746acaa","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-03 18:30:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa05d701a7ba554544cff688bb658bbb6a3a8ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:36:40.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/python-glanceclient/+/969491 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"64253f457ec21404362100b0c3ac5be9595ab635","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2025-12-05 10:49:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"14746a34f32226d4cda2c416f68b6581b0319bf3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-05 10:49:27.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2633/.","accounts_in_message":[],"_revision_number":3},{"id":"6a21c012fc8eff4672c5ab7ee1104ebb6b949226","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-05 10:49:35.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2781/.","accounts_in_message":[],"_revision_number":3},{"id":"9fa7d6e753aae1c9aa7927c2f72d0dacac94300e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-05 11:06:32.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/132/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ca63dd489223c9b45d9d7430aa4d0c6b0961581e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:22:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20fbea0a3e144d0db1a1ccc0ed76b50d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a3ae1c2a5de4ceaa27bea40dd7fe59a : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4b52aecd7ae411ebd4afdb723c65e0c : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc01d306950b45e09e3c1bc4acdc31ef : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51f88faf87e80e3c101deb33b1a09102e19da7a8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-05 11:45:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d879f839a8c545a4aea4dce9f2f4d3a7 : SUCCESS in 52m 24s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/82dbd7cf61bd453690b767012cb9499d : FAILURE in 51m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9969ee92604ac3c85e722540380024f1de4a2bd8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-05 11:50:28.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Idbe38bd27fa213069ee26305e5f519b9f5dfcb38 .","accounts_in_message":[],"_revision_number":3},{"id":"78fcc632b1a32544f4548ee714d4d91eefe92510","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-05 11:59:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/969493 : FAILURE in 1h 07m 31s","accounts_in_message":[],"_revision_number":3},{"id":"65df46390db411e3eac3c59e65d640d0afd80f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 13:19:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b51a6877b3a4868b7ba2aae47f24522\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3725628680e04541ba9f5cd868d1a9f9 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6daf509969174d1fa75a8bdd91690e5f : SUCCESS in 13m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e48037cb951c4d87b1d9f89da6ca1120 : SUCCESS in 14m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de9ea1c2b294dcc890c7ef8b48cb4f6 : SUCCESS in 16m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/4db8f00775ee4dce913989a39f37b958 : SUCCESS in 51m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/75bb269208f54a679a2cd44814f49a94 : SUCCESS in 50m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f91110ff47654690836e99e5750c83ae : SUCCESS in 1h 37m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f302775ee404685bd70e2899a293585 : FAILURE in 54m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc943ebec0e243cc872e4948349a06e8 : SUCCESS in 16m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d3a27d4669d44bc194268647a1adcbfc : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/39dcb0fb79164ea0ab506d0a1133c26c : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e2d6a56d858447ddb5c3485ed95a10bc : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/591fbb56068b4dec903c106fd78e8896 : SUCCESS in 7m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/60734d4f639b4aaebbc4472efbdfcb01 : SUCCESS in 8m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bff32a11141f41eb9eb30f7070e024e1 : FAILURE in 32m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6706e2fe1ec0404c84904a35e0fee660 : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c94d7d9d2cfa45748a75af2ce554cbd2 : SUCCESS in 1h 46m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4a23b23095ee4a0c8a32745990287a31 : FAILURE in 2h 25m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5813cce0158f42d4b6f1dd261b27282a : SUCCESS in 2h 01m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5c2cd11d553a4e20aa4b0641e90fd53c : SUCCESS in 1h 49m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/83c0529baf764aac9482f8e35b7a986e : SUCCESS in 42m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5981fbc211354a8f8a326f8b8d111ca3 : SUCCESS in 36m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e171369a77a84218b9a20d92496d81d9 : SUCCESS in 1h 38m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c64c1fa4f1a470081fade6dfc8d3d56 : SUCCESS in 44m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8533e92fb2f74453a51cbfb9cbac1f69 : SUCCESS in 2h 04m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/20f79df486e34200abfb1159f1ffe4d3 : SUCCESS in 2h 21m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bf6340fdd4842549db0eb7b66058690 : SUCCESS in 1h 43m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46863016a13e4b6983dfaf3ac2d27c41 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":3},{"id":"65c6f421e1cd5a7382ff38ce42365759a2bd890f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-12 19:48:36.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/969493/3/20251213001834/NEC-V-ISCSI : FAILURE in 3h 34m 16s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/969493/3/20251213001834/NEC-V-FC : FAILURE in 4h 29m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ee2eb43b053fbc0ad7b16b080544715553240b45","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-13 00:55:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-93-969493-3-fc : SUCCESS in 1h 3m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-93-969493-3-iscsi : SUCCESS in 1h 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b9e2115a716d366d7499b32d0c29cc45710e5051","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-12-13 18:36:47.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/969493/3/20251213172059/NEC-ISCSI : FAILURE in 9h 33m 52s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/969493/3/20251213172059/NEC-FC : FAILURE in 10h 15m 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21661a4916c8b184f67fc6c7c7db74f597a9822f","revisions":{"46e3c58d21d317139f86be78a8a9744495b07c42":{"kind":"REWORK","_number":1,"created":"2025-12-03 16:45:09.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/93/969493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/969493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/969493/1"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-11-11 16:35:44.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-12-03 16:41:12.000000000","tz":60},"subject":"feat: use image multipart download","message":"feat: use image multipart download\n\nThis introduces a new config option:\n\u0027glance_use_image_multipart_download\u0027\n\nAfter enabling this, cinder will try to download images from glance\nwith multiple threads/streams in parralel.\n\nIt can improve overall create volume from image speeds significantly.\nThis can be useful in situations were the glance backend has support for\nrange requests and has limited singlestream performance.\n\nDepends-On: Iee5c385742b4fdb0e72ef6dc0c874f7b1da2e141\nChange-Id: Idbe38bd27fa213069ee26305e5f519b9f5dfcb38\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/46e3c58d21d317139f86be78a8a9744495b07c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/46e3c58d21d317139f86be78a8a9744495b07c42"}]},"branch":"refs/heads/master"},"b277adaf73ef1ea294aba42e4a61c06e6bfcefe8":{"kind":"REWORK","_number":2,"created":"2025-12-03 16:49:07.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/93/969493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/969493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/969493/2"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-11-11 16:35:44.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-12-03 16:49:00.000000000","tz":60},"subject":"feat: use image multipart download","message":"feat: use image multipart download\n\nThis introduces a new config option:\n\u0027glance_use_image_multipart_download\u0027\n\nAfter enabling this, cinder will try to download images from glance\nwith multiple threads/streams in parralel.\n\nIt can improve overall create volume from image speeds significantly.\nThis can be useful in situations were the glance backend has support for\nrange requests and has limited singlestream performance.\n\nDepends-On: Iee5c385742b4fdb0e72ef6dc0c874f7b1da2e141\nChange-Id: Idbe38bd27fa213069ee26305e5f519b9f5dfcb38\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b277adaf73ef1ea294aba42e4a61c06e6bfcefe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b277adaf73ef1ea294aba42e4a61c06e6bfcefe8"}]},"branch":"refs/heads/master"},"21661a4916c8b184f67fc6c7c7db74f597a9822f":{"kind":"REWORK","_number":3,"created":"2025-12-05 10:49:10.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/93/969493/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/969493/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/969493/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/969493/3"}}},"commit":{"parents":[{"commit":"5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23","subject":"Merge \"Hitachi: support extending volume has snapshot(s)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bc8ed7d63c53be56ac3ed245f3c94a463fb8e23"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-11-11 16:35:44.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2025-12-04 17:42:01.000000000","tz":60},"subject":"feat: use image multipart download","message":"feat: use image multipart download\n\nThis introduces a new config option:\n\u0027glance_use_image_multipart_download\u0027\n\nAfter enabling this, cinder will try to download images from glance\nwith multiple threads/streams in parralel.\n\nIt can improve overall create volume from image speeds significantly.\nThis can be useful in situations were the glance backend has support for\nrange requests and has limited singlestream performance.\n\nDepends-On: Iee5c385742b4fdb0e72ef6dc0c874f7b1da2e141\nChange-Id: Idbe38bd27fa213069ee26305e5f519b9f5dfcb38\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21661a4916c8b184f67fc6c7c7db74f597a9822f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/21661a4916c8b184f67fc6c7c7db74f597a9822f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
