)]}'
{"id":"openstack%2Fcinder~969541","triplet_id":"openstack%2Fcinder~master~I83cadafdacd4569e9ae827b108b1f3195f1127bc","project":"openstack/cinder","branch":"master","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-12-12 15:18:31.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I83cadafdacd4569e9ae827b108b1f3195f1127bc","subject":"WIP: NetApp: Fix setting maxiops and miniops in QoS spec","status":"NEW","created":"2025-12-03 23:13:40.000000000","updated":"2025-12-13 08:20:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"beb18c981f047712701ab7a7b1202831df3c53e8","_number":969541,"virtual_id_number":969541,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-04 01:31:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 23:13:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-12-03 23:32:52.000000000","updated_by":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"reviewer":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"state":"CC"},{"updated":"2025-12-03 23:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-04 00:13:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-04 00:13:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-04 00:21:07.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2025-12-04 01:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 07:59:54.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"CC"},{"updated":"2025-12-04 08:00:23.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2025-12-12 15:18:31.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2025-12-12 21:48:59.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2025-12-13 08:20:55.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"49d0dfb26985393882ba051bfc2341acd2b21882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-12-03 23:13:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c817c1643f13749ee72f261187eb9ec67a5ac479","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 23:13:55.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/2623/.","accounts_in_message":[],"_revision_number":1},{"id":"a79308e4532f27cfa80faddcdac663479a098c6d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-12-03 23:14:01.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/2771/.","accounts_in_message":[],"_revision_number":1},{"id":"773e78e2a91b471a7da00818af5b0e9b9a329735","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38568,"name":"Xue.MingZhong","display_name":"ZTE storage CI","email":"coenocyte@hotmail.com","username":"xuemzh"},"date":"2025-12-03 23:32:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nhttp://129.11.11.11:8085/job/cinder_for_zte_usp_dirver/120/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1a9970737579fb071282b0d43ed0bb873da2de2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 23:35:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70ee42a8c0b4e10a0a503e6e2ac8904\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/453b50c97e154b3883e06102082ae96a : FAILURE in 19m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/83a553eebd5c4f8e8bb6274d70bf02c2 : FAILURE in 12m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa6b209f8f654c99932efd177dec28d1 : FAILURE in 15m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d663eddb479d5488096ba0c449ca13874b0981b4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-12-04 00:13:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/eac9ad0a2863419687dff6b592254b3c : SUCCESS in 51m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c112676f7f0145a6981a2df285d2c567 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"73c45aed00b5737333b1e7656810c9d885bf7d70","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-04 00:13:31.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I83cadafdacd4569e9ae827b108b1f3195f1127bc .","accounts_in_message":[],"_revision_number":1},{"id":"a81479c773bf31b4a6b42393adec5d61c93caf22","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-12-04 00:21:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/969541 : FAILURE in 1h 06m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3dd41e8433db851527a9bf92e072518d384ef51b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 01:31:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55266bb43c954dd3b475e2d224ef03f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58de67ae91144164994b73fc9090c22b : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1438ee33c16a47ac97142efc1910c905 : FAILURE in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31c173540aad4298888a4a4f70f5d81d : FAILURE in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e57b5897fe4fa88e380f9cf4c1b0a6 : SUCCESS in 14m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4156eae5f8b64c8d9bf3d7686b99a0c4 : SUCCESS in 45m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7553a11c09d54a8cada92ac2b975497f : SUCCESS in 57m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/312be212308d489d8000b7dc69637a31 : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6af6374997e7471a9e934e7b6f2bd7cf : SUCCESS in 1h 03m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/659554d10ea84d81ad97d92e08584449 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d6a4cdf72ba14293a5958ed4ce74587d : SUCCESS in 7m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc735caa0c7d4c508f7a5df8b2db3ff5 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b9a69b6fe6f430d837afd5d4e4944c9 : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c5816a9335a495b8b12b35ecb23fef7 : SUCCESS in 8m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f0ba5cf49f943caa9667a63169da4a0 : FAILURE in 20m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f29ae07b8cc41d2a9d288572e553cdf : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/959636f754dd4509a72a6e6850400c24 : SUCCESS in 2h 00m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/343601ed1c554cb7b0ac42148da68053 : FAILURE in 1h 44m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f4b6fdde0394bcab7564e53b2b8bca6 : SUCCESS in 1h 56m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8f6adcea8d7147f8a3981fa0f0ca893e : SUCCESS in 1h 54m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fc8d0e06ae0f473a9e288c09c2408ded : SUCCESS in 35m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9a5a509f46254fcda01eae1f319389ab : SUCCESS in 1h 14m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60b005404d6f4889a4b1f96c11722778 : SUCCESS in 1h 29m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5264ebc71b7740569a80e006c6c84e0d : SUCCESS in 41m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fa200d33e2924390a8506d781d4a6254 : SUCCESS in 2h 12m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/452c2854c8584d7ca24b128f17bda3e5 : SUCCESS in 2h 00m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cc01d87591144515a5fee51c4ae0ac47 : SUCCESS in 1h 15m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08ca075b433b487cab81d9019f8b6fb8 : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"abeedf3c7c5586c457b3d7e93db38b2aedade262","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-12-04 07:59:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bec7bda1503fe399d4a5e95f02882389c6cd4945","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2025-12-04 08:00:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e3521e129ab8f97f3f40ac846c7b5610fd7019e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2025-12-12 15:18:31.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/41/969541/1/20251212195240/NEC-V-ISCSI : FAILURE in 3h 34m 43s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/41/969541/1/20251212195240/NEC-V-FC : FAILURE in 4h 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"72946d40d3e8408da16fc51114c363fb1efc863b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2025-12-12 21:48:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-41-969541-1-fc : SUCCESS in 1h 2m 38s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-41-969541-1-iscsi : SUCCESS in 1h 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"beb18c981f047712701ab7a7b1202831df3c53e8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-12-13 08:20:55.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/41/969541/1/20251213071013/NEC-ISCSI : FAILURE in 9h 54m 36s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/41/969541/1/20251213071013/NEC-FC : FAILURE in 10h 10m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ab9847bdba042388b4a755fecf519c0a6f94c5a","revisions":{"6ab9847bdba042388b4a755fecf519c0a6f94c5a":{"kind":"REWORK","_number":1,"created":"2025-12-03 23:13:40.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/41/969541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/41/969541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/969541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/969541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/41/969541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/41/969541/1"}}},"commit":{"parents":[{"commit":"f38c2950b13a2cde2ae9f146cf525a9d54c8b830","subject":"Merge \"image_utils: Detect missing device before calling qemu-img convert\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f38c2950b13a2cde2ae9f146cf525a9d54c8b830"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-12-03 23:08:40.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-12-03 23:12:32.000000000","tz":0},"subject":"WIP: NetApp: Fix setting maxiops and miniops in QoS spec","message":"WIP: NetApp: Fix setting maxiops and miniops in QoS spec\n\nCurrently the NetApp driver compares max_throughput and min_throughput\nwhich are in the form \u00271000iops\u0027 and \u00275iops\u0027 respectively.\nWe should be comparing the iops values in QoS spec which this patch\naims to implement.\n\nCloses-Bug: #2133831\n\nChange-Id: I83cadafdacd4569e9ae827b108b1f3195f1127bc\nSigned-off-by: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ab9847bdba042388b4a755fecf519c0a6f94c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ab9847bdba042388b4a755fecf519c0a6f94c5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
