)]}'
{"id":"openstack%2Fcinder~973592","triplet_id":"openstack%2Fcinder~master~I45d32892e7712cb3bb4e52cf4547691505b01d01","project":"openstack/cinder","branch":"master","attention_set":{"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2026-10-06 11:12:35.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-10-01 08:54:24.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-10-06 14:17:38.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}}},"hashtags":[],"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","status":"NEW","created":"2026-01-16 10:04:44.000000000","updated":"2026-10-07 14:42:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"1dcc813852f276d5f5b759991a5e4a4e87db1aa6","_number":973592,"virtual_id_number":973592,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-06 13:19:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":-1,"date":"2026-10-06 14:17:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 10:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-16 11:07:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-16 11:46:30.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-16 12:04:30.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-16 12:19:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 12:26:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-16 13:25:02.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-01-16 15:25:36.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-16 19:20:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-18 12:19:35.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-01-21 12:52:06.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-04-02 09:34:59.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-02 10:07:23.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-02 15:23:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-04-03 03:06:11.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-04-09 05:31:17.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-25 07:48:30.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-06-19 19:51:30.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-09-30 14:54:55.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-10-06 12:14:37.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"CC"}],"messages":[{"id":"0bf877621f40b9c82d00d5681ebfad7e14a05130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-01-16 10:04:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de975ef8d1261cf26ab464fb72e30569e4bb8554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 10:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619496bff6e14cd89764a283dd0e5e45\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aad3697a3eb74ba988d6e024e8690696 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e6c378849f3459488db9d923aacd50f : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e93c549cbfdd4eb6b844148b8316bbe2 : SUCCESS in 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2284870133c86206dda5270583f4bcf7f0e6a303","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-16 11:07:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":1},{"id":"efbf932811716e04afbf34d1949f6f29a9559d6e","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-16 11:46:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/48547621ba4540e4b9f655b59d83234e : FAILURE in 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d3921018c3b1668358c115cd60a7dfb31cdacf3c","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-16 12:04:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3661d43544ef4d9b9cb511b24eae497f : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7a5a8d86343e0d4d3bcef4db6aa6b3b21d8de9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:19:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffb99df4cf4241f1b17de94265315136\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd682543a5e4554b091e1e8bcecdea8 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f421157d2a8f4f19ba4278d2d9903894 : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27ab637709c8467b93b14714f6cc87ed : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea5c974a6b1a4ca4a4ef722c7904f4e0 : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4c768496ee8b471290f22f44fbde4aa1 : SUCCESS in 30m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16d257e3937b4d679b70f2c157a228de : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a307c62f4a6941629c42ca44dfb41cb6 : SUCCESS in 1h 25m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f263660e18a48848845f1b3dfe2780a : SUCCESS in 1h 05m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ddad5ff2de1a4b57a3dc0029a06e63ac : SUCCESS in 17m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aae0ac8b7fc84631a7441e32f2733377 : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea78673e8ed848bdbd09bdd5c1e2810d : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ff5015cebe5409dbb8a54bb1a8f34d7 : SUCCESS in 7m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3ca8ee4d7c2843c4a27dd6a80c384bfa : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39262c0d17874362af6682d6466475e8 : FAILURE in 18m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e61923a6300241fcbcf740ec9b5fae5a : SUCCESS in 4m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b0dea9aa2dd49389914817e859b8fb1 : SUCCESS in 1h 47m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17033bf9d97946feb3b000f24e1d429c : FAILURE in 1h 33m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7dacd9fcf01e421587b48720a6630ab5 : SUCCESS in 1h 30m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6fb839a01c91418bbd1171f7f9bb9412 : SUCCESS in 2h 08m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ac869cbbb66549d3aece8523ee2ca7dd : SUCCESS in 36m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc4645d43d324bc5aa862d47388516b5 : SUCCESS in 1h 12m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/192de489cf4e4087a9b580bedb1325fe : SUCCESS in 1h 29m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdb04e175a6645dcacdb6c87b0138ff1 : SUCCESS in 29m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27dda4fe910f4bf49b15383178befa1f : SUCCESS in 1h 17m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9d977dd85b4f41f780884ba77034d710 : SUCCESS in 1h 30m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54c27130809a41a98a388454f754f64c : SUCCESS in 1h 02m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70c0ef605f6246b8958fdd948df30c4e : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9f3d35ce68a763b2cee17eb53e59f62e9c45ff30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-16 12:26:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_01_2026_12_19_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"7f89da2fbac624d672b4aaeaf4c965ac7948a905","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-16 13:25:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/973592 : FAILURE in 34m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6cb1704892fa25652f2797356677529a0dfdd2fb","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-16 15:25:36.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_242/index.html : FAILURE in 82m","accounts_in_message":[],"_revision_number":1},{"id":"9d19234410110a81135f89c52e4e98d05d35a081","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-16 19:20:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4350d2326dcd4c01abd645a5b9bbeb06 : FAILURE in 59m 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8388b16d5185492284435f1dc8a0f3ef : RETRY_LIMIT in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"64fb5a2de3f18a127dcdf201e8cf2b2c7e39df11","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-18 12:19:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-973592-1-fc : SUCCESS in 1h 28m 19s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-973592-1-iscsi : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7a6cc135361384c35afb1b550f649d5b9124ef52","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-21 12:52:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17368d0b271977a0a94b45319f79ed45fa5f8d3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-04-02 07:57:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d6fbe92b45447b03823fc6340e763060f48717f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 08:17:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a8f46c78badd4e2990bb57f2681fd5d1 : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14a521f3bc1a4b1ebcb04bf667e25601 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":2},{"id":"3d2c31c92fe33488a76382f66568dbd06759f8dd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 08:38:22.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":2},{"id":"0419953e6125b0cd97efdf3214f633117bc611d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 09:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ad9486ee1d4bd68594e0d9f7300913\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/210df5863d484adf8ceeb26be9d37277 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ddf8a086acf54060ab5ab3659b564c95 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c21ef2a51b674be8af2c1b7336690ed3 : SUCCESS in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a4a18a9a0f3b870ebf522e3e4f9fb91e01025b6","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-02 09:31:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7023d9f2b45244b98ae91905a09da89f : FAILURE in 24m 17s","accounts_in_message":[],"_revision_number":2},{"id":"62e6d1d6139fc12c031de006b4dbdb8863c5b1ef","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-02 09:34:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/fa4d7338bff640568aa0144768899a8a : FAILURE in 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"dc5e51216e5498502799a496fe54b5059a56a474","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-02 09:59:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/973592 : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f17c3b679e7ef80b7a0cf302e99af909d735f01c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-02 10:07:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/5694fcae9cb34bb9ac1a712198e0f49e : FAILURE in 36m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f49e37784e034d489f7e0f9538b094330753c41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 13:23:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a39736efde4a819bcfc0a58e897ca9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08e28c8113c4a0da6aede60afc2eaa4 : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bf47254223941b1aafd3b02d83dae08 : SUCCESS in 7m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e66ba3f921014dc4aeb9a84914441295 : SUCCESS in 11m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/232c05791eb44addbe7bc3a958e898f0 : SUCCESS in 16m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0f8eb3bc0a44838e5152b3a0563cac : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/61abec61e9e1458588595ac4bcba54a6 : SUCCESS in 1h 08m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b525ebce47e4622903e53ccdc80a66d : SUCCESS in 1h 06m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/635653e8cf0b43f6897378ee3584c9f2 : SUCCESS in 1h 52m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d0097b6d29c408aaf55a975d1934677 : SUCCESS in 56m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1b7309ea9da34e9784b7ccac14f59ff7 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f7be4d65901455caa3465ede8f4f280 : SUCCESS in 2m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb086bba90c64784b7d8a4687b854647 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f0fbf5052b94e1e937750b00e5eda44 : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/51443497ad1e4ce089a72145aa494da9 : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0951e0b3b5240368bb82d5180729e7b : FAILURE in 28m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c818c7569dd7479ba0eda86d0d4a1c3a : SUCCESS in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/686d9962233e4f61a548b4f370afcbfd : SUCCESS in 1h 02m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31f33361a79e474ab38857cf4318e003 : FAILURE in 1h 37m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94d3d8a18b0c4ee9bfe8cf5bd7fac60a : SUCCESS in 1h 55m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8dce408b6b274bf1b59d7618bef25419 : SUCCESS in 2h 19m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3277f105768a44a789e346f16aa02225 : SUCCESS in 25m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/834267e621984a9384811dfe52c87008 : SUCCESS in 39m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c89312c6daf34b6697b712a95c8933c4 : SUCCESS in 1h 36m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/119d253686cd4fa78dd6780ed15e7189 : SUCCESS in 36m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4525abcc744f4002aab0c5b1e5889d2a : SUCCESS in 1h 15m 44s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/cb41887e60024897846d18df9671c09f : FAILURE in 1h 29m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b3c4d4379214916ab86ad7ccff855e2 : SUCCESS in 2h 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a515f1d4e7644cddb7d8b30bc0104cc7 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c1d9fca47593fa5db456fa5a13914b9c8dee0e2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-02 13:32:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_04_2026_13_26_29\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"94573493e45e564bac80d0e791a1c13151167127","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 15:23:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3680/.","accounts_in_message":[],"_revision_number":2},{"id":"2bda64e0ad7aaa60226538f15c18cfaf305b1fd0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-04-03 03:06:11.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-973592-2/FC/ : FAILURE in 19h 8m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-973592-2/iSCSI/ : SUCCESS in 2h 47m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f2ac990f98bbb19e94f84f399bcd4f04141838b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-05 09:07:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3504/.","accounts_in_message":[],"_revision_number":2},{"id":"6101d388150eaee0c6d9a35fa0de8c3053d982f1","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-09 05:31:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eebc9edf265ce88497e75dbd4e78ad298dd1f9f0","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-25 07:48:30.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/973592 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"1748ff9dcd451fd255c6cbc8feeeb52ee3e456d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:51:30.000000000","message":"Patch Set 2:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1333/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"d2c0169bf719d1e67826981a641a328f11099cf3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-09-25 07:59:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"b037ef68d3b3e8cb851e8161ce466d3edd6a6e4c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-09-30 14:22:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b0f250e91985c07277af0a7a9646064a465a13db","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 14:54:55.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/6a5fc17afa8c46de8e68ef37c05122c0/","accounts_in_message":[],"_revision_number":3},{"id":"e743541d833a66041057490f5e8495f8537b83a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 15:02:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9430d07a87b4749bf77660a1323b12f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9a263eeb2584acbac9aac621b6c2616 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aac1eae8747e40c284bbbc8e3ed93caf : SUCCESS in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad7e7fa88a323fd32394af5fda44e503d309b322","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 15:10:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6a5fc17afa8c46de8e68ef37c05122c0\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/557831e4f5a046839b6a9cbbe1a7e00e : SUCCESS in 39m 20s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c333c47d6c44d528e8ec9c2db095730 : SUCCESS in 40m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/36792b47bea74c32b8f84770e7ac0eb5 : SUCCESS in 43m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb40aa7233f7482c9010abb347c6a1f1 : SUCCESS in 28m 05s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cda2bf18a124ec88774ddffb3c69585 : FAILURE in 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0d3bb151c46417082648022517fcb8393cc5a3ab","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-30 15:22:30.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":3},{"id":"1b988d731de97023aed7c115cceeb897ae825c42","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-30 15:31:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8b0fcac6c79c4c0fa8c10fe74d63bea0 : RETRY_LIMIT in 16m 06s\n- cinder-storpool-tempest-iscsi-multipath build/84186e6470034bf396d6c90882beee42 : RETRY_LIMIT in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e961e16e65727493b476d2ee76de393718227d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 16:42:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/929f296af48f4a06b1adc79c70295c60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5379c6f71ae142d2b2efbb090b28ad9f : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84a280b5fe2f4b2c9b49e3b91d5374dc : SUCCESS in 12m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2da7854dec5e4de89fc5e5430d86dfef : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26080906d7d942d292604a20c0e258b4 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/a8800c30f7f140cc9287052e82177b25 : SUCCESS in 1h 14m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/090776d62fa44c969589b55979960964 : SUCCESS in 1h 16m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d55c4329d91c4d8190ffe46da8b3734d : SUCCESS in 1h 17m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21aa11bdef1e4db9819a23a84f1e3445 : SUCCESS in 1h 02m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c1bfbf47b7420a94368968fd16580e : SUCCESS in 6m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3cfef48fead14e70b112dcf08a781e2a : TIMED_OUT in 41m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3a1906524354ea2abcf27b8a941adec : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4e4c8d07540489691d1901b308f7727 : SUCCESS in 6m 41s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2f8e172dd9c34951af67880581337297 : FAILURE in 12m 46s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/070f8ece243842998906c88562983dda : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3d71a2b558e64ef99287a2b948bc2e65 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/31dafe40f691400fbefd2d030a0cc7ca : SUCCESS in 6m 20s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/be8225afe731471c9f8446a1a8b7b5d7 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b2b3f050eda5449eaeb67d098cafb9cc : SUCCESS in 10m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7f9a21a3afcb49788d04147d0b665730 : SUCCESS in 39m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bd1c22def89b4904942b7b1fd4dd63d0 : FAILURE in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/babb79c0696d4d46ab6fc386e00f6164 : SUCCESS in 55m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/370c32fa2be34952a0f2a55f8cb9a99b : FAILURE in 1h 53m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/212390f2eca64cc983cb1189030f8040 : SUCCESS in 1h 40m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5e294248084649ba8445522d145de407 : FAILURE in 2h 05m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/246b884f508045abb8df1a08799348a5 : SUCCESS in 1h 53m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cb28e357e540460b898a9f126cbd57cb : SUCCESS in 38m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/939fa4908f924988b618517ff006835c : SUCCESS in 1h 21m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bd02fa0c361b4512a551d8fcfddaebcd : SUCCESS in 1h 42m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ec8cf591f0d145f7b2e4f6d05f5cb8f9 : SUCCESS in 32m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd0bb59333994fc98a7f76f50826bd48 : FAILURE in 1h 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae722662580b47ba884517e4ab985a33 : SUCCESS in 48m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"d0bc6e181c61792e051fe061fad0486b5a37fea9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-30 17:09:49.000000000","message":"Patch Set 3:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : SUCCESS in 0h 48m 39s\n- FC       : FAILURE in 0h 8m 38s\n- NVMe/TCP : SUCCESS in 0h 47m 18s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":3},{"id":"155cb5ead138dd300463df0a1de519b1a34b24a3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:53:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"06c48c4c28855bfa443ad7af720ef102bf25da1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:54:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"25ecc7f9c1c6c8140fe815f3a1586917bbcd125d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:54:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"6a6780ff62f6e1d4afc333b33a316e6828f42f5b","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:55:30.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"74802e0ece41f8b72e8d2b22d0471b68060cb949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:59:33.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9de42875df35730dfc1f1f53a3a9d6a9031b2788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 09:00:25.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"14a40e2f5f75d24e38e0af43171acc519a0b7f7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 09:18:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d679e0c654e44fea906a11436b80028\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa4997811f784b968b999d1ac0fc62ee : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b13954ab11eb4c31aa138d9e47b3300f : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d38e30eaed47080adbc0869082cd14f2dfa7e70f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-01 09:18:18.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f535ae7e0d5f4b09a6f364d0b1c905a9/","accounts_in_message":[],"_revision_number":6},{"id":"eb902a47a590aadfce95388c79d5d636f7a66108","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 09:29:14.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5777/.","accounts_in_message":[],"_revision_number":3},{"id":"23e75d943e80bb1eb155dc8fa5c2dcbdb9809707","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-01 09:45:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f535ae7e0d5f4b09a6f364d0b1c905a9\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24fa0e2c97134132a808d08b74edaac0 : SUCCESS in 38m 52s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/715e7d3d7d5d4a98b71de70839279a97 : SUCCESS in 34m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83a981f1fdc54c8089a7e35e19684aa1 : SUCCESS in 41m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb8d563fafeb472d9e4900e7cd63bf16 : SUCCESS in 31m 17s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/67117d94ecf241498820248fa953f581 : FAILURE in 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"edcd1f570f857f5b64ed0900879c4dfac42cd68d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-01 09:49:33.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/382d0e402cb84e77bf3fcab80604f34a : SUCCESS in 46m 02s\n- cinder-storpool-tempest-iscsi-multipath build/fddb46eb9eb545078550f1ba03f6c2b0 : RETRY_LIMIT in 5m 20s","accounts_in_message":[],"_revision_number":6},{"id":"36e80bb2897057ac8dc7fa1229ab452b6437f6fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-01 09:56:58.000000000","message":"Patch Set 6:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 53s\n- FC       : SUCCESS in 0h 46m 47s\n- NVMe/TCP : SUCCESS in 0h 46m 33s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":6},{"id":"e99eab90eb357022abb7f77d5cc3f5b1037230af","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 09:57:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":4},{"id":"7b61c7fa3a618f83683b57d4fd1867088a10179e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 09:59:22.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":5},{"id":"0f4146df86bbfbfd875b1fd8f1c8100a20ab5d71","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 11:00:02.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":6},{"id":"33458148f0abe22921f77367f13ff929a130cb2b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:17:42.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5600/.","accounts_in_message":[],"_revision_number":3},{"id":"6abeb8acbe0163b467ceb548f17a2f1f58c4aaaa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:19:25.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5611/.","accounts_in_message":[],"_revision_number":4},{"id":"ffd62fc7971758ef7133723aa7e2711dbcdcc761","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:20:14.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5616/.","accounts_in_message":[],"_revision_number":5},{"id":"91a134627128eb818b58169119db44b2c1d05f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:24:48.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7eb87be2c94f4d368c9de39c7e4445e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66fbed4bc25c4cdf9abdab865def090b : FAILURE in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/524796ea523d42d590ebe7d1ee475bcc : SUCCESS in 12m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db1d417d51a443fabe4db521939ac1fc : SUCCESS in 16m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648c4d8e29b4490dbc9828e576715320 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b5a7a93938014b4d8d17b1c1234d2d18 : SUCCESS in 59m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49e4b73d13ab4b3faf88d8d10fbaa70e : SUCCESS in 1h 09m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4415f8ec6e7a4a8d8c514995e3fb1694 : SUCCESS in 1h 25m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1ad06be6da44b699842938e0cc6dce1 : SUCCESS in 57m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/211c98497b0e48b882128edb8e1f8970 : SUCCESS in 8m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4bf0b03866a84142875185e74bfcd514 : FAILURE in 24m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5f155dce683045f38648398352d48261 : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f942922fc1a4236be32fa674a3f9dbf : SUCCESS in 6m 48s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/5fc1eee1ec6a4d978cf3ad0f25b9a9ae : SUCCESS in 11m 40s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/53e8ab5dc11842ed8264a80e865e24a0 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e40fb289f0774e5abce5e60fd3344f8c : SUCCESS in 7m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/11edf52295c547a4b2c8c7228fff4d9e : SUCCESS in 7m 57s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/43a4c23f68f54cb09a0884e6ca52fb2f : FAILURE in 6m 09s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ea6ec53a08fb426b8bc17289a1438206 : SUCCESS in 10m 30s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1bc8c820312443f89a5ad6c6deb9fec2 : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1aef982f0af44f8b698b6f1a411b090 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ded162ebe0ce411b8d2da4021b6a115d : SUCCESS in 1h 39m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ceb95479faff4e2ba2e3647f2ad33e8b : FAILURE in 2h 16m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7e1c36c17f04425bc3b44ad3a50f0e1 : SUCCESS in 1h 43m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bf20cea7dc7642008dbce019b87dcb1f : FAILURE in 2h 23m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b7a7c3a0308a42a4a733c6174c58fcc8 : SUCCESS in 1h 25m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0f25556fcf08414d81236db1a5ac6578 : SUCCESS in 37m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a27031db6ef84011b79e51ace1de11d2 : SUCCESS in 1h 16m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/471dd3ae248c4b87b0916b96568336da : SUCCESS in 1h 28m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/05323d1f8b36442db93c2e9c2fd3fc99 : FAILURE in 44m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b499f09f9864be0ba01601858dced68 : SUCCESS in 1h 43m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d108dcf3e36648378783e47b328f68ec : SUCCESS in 1h 06m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"d1b466338d475049a8c73a589b6734c49c7536e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:51:35.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5621/.","accounts_in_message":[],"_revision_number":6},{"id":"7660a26e23393178767e34cf3ebddf3a1b67aa68","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 17:51:24.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5788/.","accounts_in_message":[],"_revision_number":4},{"id":"86dd3628a087359e7460c1aad549ae1731f7d72c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 03:44:03.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5793/.","accounts_in_message":[],"_revision_number":5},{"id":"b7e030de802c286e3a8ed9585b44f3ec560c4143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 07:50:07.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d0fc2967c3c69c37808dd8873e960790d60c89e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:07:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c308dbdda88e428f8c64e8cc6ef07fad\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5037e25229f345d9a0c075129ff6daf3 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/389fc9bf0eba443fb3ada8cd972d8748 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"82573fc9c7954bdb6337a551519545918ddcd994","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 08:08:14.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/101897716ad04cffbe84d759aa4b0710/","accounts_in_message":[],"_revision_number":7},{"id":"8b5542c989392fdfce29d0f4fafb7db695c3dd72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 08:28:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c06a79975c83328a9512d695ae08af11c397c074","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 08:43:32.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/33e4477ccfad42ba80e7d703d237fd9e/","accounts_in_message":[],"_revision_number":8},{"id":"08f3ba3d7a8fb4c4b031754dd7aa2dd4ccfc5b4a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 08:44:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5626/.","accounts_in_message":[],"_revision_number":7},{"id":"ea7b401216c8a900b31b6b58f88c8c6efd6a4c1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 08:47:58.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":7},{"id":"f7b378c522ab37c0d5a555312c9e0241087286aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:48:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c323d2039a0a4eb38bc6d25003d6e4ca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/536008e4f7d9411fa20651b934f9db4b : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/85b16196503c4fbfb7d296601daf1bec : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d21288544c831f338a2b84e9284938ea4aad0b09","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 09:14:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/33e4477ccfad42ba80e7d703d237fd9e\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e42896cfbcda4627877a8db3f8911de8 : SUCCESS in 37m 03s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0a67e400244440799db29e550d5aa600 : SUCCESS in 34m 10s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5adbcfebe0924118964671fb26bbc37a : SUCCESS in 40m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5d94b0d929b4a54bdafcc472129aa6e : SUCCESS in 31m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4581dc7741fe42a0b345ae2b67847264 : FAILURE in 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"33efec664d1e9002b70dd12fa58c5c49ea6651e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 09:21:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d0438ecefa9c4995a89947042812d790 : SUCCESS in 49m 49s\n- cinder-storpool-tempest-iscsi-multipath build/c7a025ae764f43a7bd31e9ac5df5b94a : RETRY_LIMIT in 5m 38s","accounts_in_message":[],"_revision_number":8},{"id":"3ee56dd74b7b18b695c76b07a97a0ca5aa7e4559","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 09:25:01.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":8},{"id":"1cc40ff8236f5ec162c03b82556452a94022dddd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-02 09:27:11.000000000","message":"Patch Set 8:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 28s\n- FC       : SUCCESS in 0h 49m 22s\n- NVMe/TCP : SUCCESS in 0h 46m 16s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":8},{"id":"d5deefceb2db77354b7d462f2b05e13adfc936fa","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 10:08:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"88fb9344e59acc45583e66e241dc55b49bf33a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 11:05:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f9cc1308f944d519bd6b10e14c7873d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15feea5c2b854e74b90dd22d9550d712 : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40a5a1b18645410c970b0d13ef0f6da8 : SUCCESS in 12m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/301f7401ff064d79830ae84a3857b5f3 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41261621dc484d958edd82856eb03182 : SUCCESS in 15m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/fa44b049f2624129873d9c2e1d80a704 : SUCCESS in 1h 06m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e7e9c8616f0476abd04532a5ac77bd5 : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54e340ab55ab4af9ae9bb1887ac9faf2 : SUCCESS in 1h 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0815da948c814ba19aacb8f0666fe591 : SUCCESS in 1h 01m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cecc09a00ac454b9f87d27487523eba : SUCCESS in 9m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce9d841103274edd89e83cca71129d2d : SUCCESS in 20m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/efa3c988d5ea4cc8ae2dfb1724490d72 : SUCCESS in 6m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e71f7e91a62f463aa540721cad6d8285 : SUCCESS in 7m 38s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8cf1748ab47e4966947584becab670f4 : SUCCESS in 12m 43s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2ef3387452734240bb2d58656638b7cd : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/62ff58fe282042ca89fd01a86fdab116 : SUCCESS in 4m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e53838f0dbe64740bacd1140e86d7e6b : SUCCESS in 9m 48s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/66aab298138a47c78564216ad3fee24d : FAILURE in 5m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c50566bdd56b498ab6029328c56be0d9 : SUCCESS in 10m 09s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02b79bc084264b6f99ea1b1b90d80369 : SUCCESS in 40m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86e72f7b27674cd2ac073e77465fc33e : FAILURE in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f002d509dc804b6f8912ffd7d827a288 : SUCCESS in 52m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a3825a5575dc403fbf7e34d94665a3de : FAILURE in 2h 20m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4cbbcfa7dd1542d0ab4563c0eed31351 : SUCCESS in 1h 43m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a0d963a09dd04c58a6b2b23e433c9cf3 : FAILURE in 2h 28m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/397e5bb682664b8b81eade35ee386883 : SUCCESS in 1h 47m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d1995304020e4b96b693ebc3b17226f9 : SUCCESS in 40m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d24ede8e2104f0c8491df245ee6f9f2 : SUCCESS in 1h 14m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac7d2e65510a4c63975a6185dc1fcd18 : SUCCESS in 1h 40m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/09edf82c3e21416a853fce4236485f64 : SUCCESS in 45m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fa021560d324c59838bc11d21f45a84 : SUCCESS in 1h 37m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17610f50d3e54db7a26fe9ac35afacf2 : SUCCESS in 30m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"c406f56339e20593a5c48fee7a9ad4418ea8703b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-02 11:24:24.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_10_2026_11_18_26\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"bc0d55e2449ef2bf0e74f919b3193e8efc648945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:03:25.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"0862407c548d171b807b02f08ddc7cc2c815b966","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 12:21:33.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0841e4803c254ad7a9dd834b0e2a7d1f/","accounts_in_message":[],"_revision_number":9},{"id":"1e66a504bbc038ea384b6f3f888cf4d032a334c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:36:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"24ebfc0d2a411649d02896b2efaeacf7ad02ca65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:41:38.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent tree, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"8858968093e0952d157510b2076ad88969fec72a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 12:49:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2cf9ec1a37a4052ac89ff5d3a661424\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9c8a5d663b5498f81b0b1ca8c39eaed : SUCCESS in 18m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/73b9263cde6b4b0eb14c446356d68731 : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1319c99ca51045d768fbb0315701c7a20bf702d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 13:05:42.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":9},{"id":"4f20fe0d2ce8044a1ed2594cb4a4d77c07fc7c8e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 13:13:42.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/791c788e910d4e85bf3b811b59e6038f/","accounts_in_message":[],"_revision_number":11},{"id":"cf009cfa8e4b917fb4848418d150ae35e1f14c9b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 13:30:09.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/791c788e910d4e85bf3b811b59e6038f\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/15d86fdd449b455f89a075c084f11617 : SUCCESS in 39m 33s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cc56c8ad0f04e8e95093393ddd3c216 : SUCCESS in 32m 49s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/be31cf1488924d98845282982d9dbdb0 : SUCCESS in 44m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/abe912c36cf74abe86572492225e9921 : SUCCESS in 30m 49s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a420217563eb400fb8aa22f5f691acf4 : FAILURE in 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f4d53f02b2752c6c4501276dfcb20555fcd3a259","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 13:44:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5631/.","accounts_in_message":[],"_revision_number":8},{"id":"612e08907e96197e361939c87329b5e90516f9f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 13:55:09.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5798/.","accounts_in_message":[],"_revision_number":6},{"id":"a979311f9261b422404f828197a1a031c49d89f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 13:58:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbab7ceb318482ba23ee235c61d3f5a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/268b785c93bc423d9875d686e2639411 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ff92dcec84a2494e825383b54514a90a : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8a27e8bd838808153b01da492efafcfb3b9e917a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 14:07:27.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":10},{"id":"10cd710e6ee31c3a718f15c2012c5e9d2f9f149b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 15:03:02.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":11},{"id":"7d36ff160aa633b7595625b95943aad158056695","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 15:20:46.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8ff93407d37b454f9b2f3967656e625a : SUCCESS in 42m 45s\n- cinder-storpool-tempest-iscsi-multipath build/f48e1fc3b2a944e99277524b938dfb73 : FAILURE in 1h 07m 15s","accounts_in_message":[],"_revision_number":11},{"id":"ebd67acff1e24cf139cd37d3734e6c788a07f808","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-02 15:30:03.000000000","message":"Patch Set 11:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 59s\n- FC       : SUCCESS in 0h 47m 59s\n- NVMe/TCP : SUCCESS in 0h 47m 14s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":11},{"id":"879e4dccc420c58201a203f4061299d6dbdfb801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 16:45:11.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c64f8bcbb53a49e9bf54a7d174d61d98\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d0d6148f4a46628b14386a3ce40e53 : FAILURE in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6367045f952e4780ace55d712f83cc81 : SUCCESS in 8m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac83cdc1489e4657844cdb8a7f2d2651 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a9b869821444db8adebd3fac1253a2 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b613fccdb2d74d90aa42446266946af2 : SUCCESS in 1h 10m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7feaa04e1abd415bbba788068b103db2 : SUCCESS in 1h 07m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/748d63b1211d4325a4ce16800cd215ee : SUCCESS in 59m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71475f9517de4d59850e4a78d5c10026 : SUCCESS in 32m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd2c571ff87435383005e30af67ca58 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f454f898156446982f0890f939d0fb8 : TIMED_OUT in 41m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4cfec978eea04194af147616b6c6fbf7 : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20a84c52be1e4b72b83820a8f59cf193 : SUCCESS in 4m 14s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d9f77756514c4eb49a6c02784732d4b8 : SUCCESS in 8m 59s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/60db5d82e737474b86ecf7d98bb9dbab : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e626d226e1cb4bc7a76c6dbfc2793868 : SUCCESS in 8m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9335ccbdfef940d2855aeb852c880789 : SUCCESS in 7m 06s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d2c2d94b4ac4464ba0dfb5d1c0f8a821 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/229ba5ec7e8a4e84a37d6d5ac91b0366 : SUCCESS in 12m 55s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3bf2508517eb4d8cbc5718558baa63b0 : SUCCESS in 39m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/195fde8fb8fd4501b16fe46e54e8e281 : FAILURE in 7m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52d2f615bef34e9590dc10aa8a728f3d : SUCCESS in 1h 53m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0df4ebef8e184df28ff85d883fd55804 : FAILURE in 1h 03m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66272dcf71574e7dab66699c92df9652 : SUCCESS in 1h 48m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b69af7fd06694339803d69b4d94a54b3 : FAILURE in 2h 08m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a7defb50d041407e80e2c6c779c7fde1 : SUCCESS in 1h 59m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dfe8167c3ce246a4a041a991a8b0169e : SUCCESS in 41m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd099583ba3c4424a950f948dca7f284 : SUCCESS in 38m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7ea91379ae004267beaf671239736508 : SUCCESS in 1h 09m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c1d00dc063cf44c39651a0856407c2c7 : SUCCESS in 34m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea42bfc49135440b9046b494b60d9dbd : SUCCESS in 1h 24m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36e68630007a43d28da5b0d40d9a40e3 : SUCCESS in 58m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"9dd07b2487639dc6bfeed0cb812efd1115fbebb0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 19:08:44.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5637/.","accounts_in_message":[],"_revision_number":9},{"id":"dd0bc2c741135eaaefcbb312ecb8e2b0d4239157","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 00:53:13.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5643/.","accounts_in_message":[],"_revision_number":10},{"id":"c23da1c4cffdcf8f3b747c56531d332073dc15b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 04:26:40.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5805/.","accounts_in_message":[],"_revision_number":7},{"id":"5aff93e90438edaa09fcc09ca346b2d90817c945","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 08:43:10.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5651/.","accounts_in_message":[],"_revision_number":11},{"id":"d6d37bc4aac3fb48d6430f9704f08096519813ec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 14:26:27.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5810/.","accounts_in_message":[],"_revision_number":8},{"id":"17c1502962fad4f43ef44a1a311f7e3fe166592f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-04 00:03:29.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5816/.","accounts_in_message":[],"_revision_number":9},{"id":"1156b2c28deeaab173f7ba55e4eaa3541a051e97","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-04 10:16:13.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5822/.","accounts_in_message":[],"_revision_number":10},{"id":"dc0c66d6fd3eb09effc8f60d6751b917e2a0f55d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-05 02:05:06.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5830/.","accounts_in_message":[],"_revision_number":11},{"id":"bc9d6bf2a2fbb5ae7cfd2429ef31671ef4872e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-05 07:04:56.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"eafc891aeaf044694e80cc2031fa5ab111519584","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 07:53:51.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ad9eca39325443a3aab491c105c59f81/","accounts_in_message":[],"_revision_number":12},{"id":"8c76d565748c82964d92f1ee00c6c3625e6dead8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-05 08:03:47.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":12},{"id":"50acd87695355a9311e433788e5ae1b88a2e2110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 08:08:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d26f45811694efdb5a487f8b99dc927\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb4af3fc94c44a938b67fce6210f2342 : SUCCESS in 15m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c9f7d7d01ac4bbea11da336220c0b28 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"68cdf2d9efed1b340f068b849af3967d18acacd8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 08:41:39.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ad9eca39325443a3aab491c105c59f81\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/819cba95d926402da6d2b9056faad8a8 : SUCCESS in 38m 05s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b1eb7be623d74ffd9b755cccc1efbd84 : SUCCESS in 41m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a458e63cf4994a5389cac6a7f9d1bd69 : SUCCESS in 46m 57s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc0431c26f93495193fc4d29aa010720 : SUCCESS in 29m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fefc98ae67434a35965b81bdb6ebc97c : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b604eb7c913aa057ec30716e30913ea9b5780d58","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-05 09:08:34.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5666/.","accounts_in_message":[],"_revision_number":12},{"id":"85a1379d4457ef59e240a46e88a9ca2d6be6e488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 09:23:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddfbf77d2a32407fb7298978a1e44c34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00ae6c69f8f4d43ac1cd44644c4e856 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65175a4f16cc4c56b6f0908ba7cf7b51 : SUCCESS in 9m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0f17cdd767e2446a8b21af248f7c612f : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee861d78cd74728bdb920b86da2ac66 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a1b06ff313944e979f86331cc87e9f4a : SUCCESS in 34m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ee98cb47a0146ae936021ff36fc7532 : SUCCESS in 56m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9394174fa13e43d5a2ffbd0b4f4339d4 : SUCCESS in 1h 23m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/642c7cc18c0e4c5ea759c665c2bd6a2b : SUCCESS in 34m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce9e4f167b4640c095d011a1a46d7b80 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8c620a3e19234f3cbf09c2b74a5857ca : SUCCESS in 25m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dba1f1803411416b8e5d90eaf5921034 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9357a5a3f7974904b8da41fd71a06428 : SUCCESS in 5m 52s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f81f07e95c804fffac31a077264bbfb0 : SUCCESS in 12m 57s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2df5b4d1d423400b8145fd28f7556a97 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/164ddbbfde06422caeb3f9fb31230737 : SUCCESS in 6m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42bc05e75ece4ee0b76dff67a9e93ac2 : SUCCESS in 5m 23s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0546fa3a24f947fdaa3261c5062bd276 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/eb01cf0ee7f44df98199756bc8b962b5 : SUCCESS in 8m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d7652b68cbed4bd9b82630dd0159be9b : SUCCESS in 42m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7e34bc4071d4ec5b6ad3cae8abaa661 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af72fc00fd3b41fb8d065219a83b6613 : SUCCESS in 1h 34m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f1e5e687ea44c82a54669b8dc8c4a34 : FAILURE in 2h 12m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/967bf5649e5a4a509d10e4cced98be28 : SUCCESS in 1h 49m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3cc67782d20c4327ad3b2aa4e6db51ee : FAILURE in 1h 55m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/492354ae497f4eecafa284301210cc13 : SUCCESS in 1h 44m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3ab71b9fb35b446fb7373a4afe8b4cef : SUCCESS in 30m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6f1e049e10aa4a17b416a7dc08e7be40 : SUCCESS in 1h 02m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22cb7bf25bdf4c57b3270cfa8723d034 : SUCCESS in 1h 34m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9a3b32c94824d45aa1044b273742d73 : SUCCESS in 28m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7b403e0268e448c9c07d5b5934d0bd3 : SUCCESS in 1h 37m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90c699f9b0f04607a0e34c05c999debf : SUCCESS in 58m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"4b4275a937da84945a926d2d6ca975105e126e09","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-05 09:44:18.000000000","message":"Patch Set 12:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 46m 11s\n- FC       : SUCCESS in 0h 48m 46s\n- NVMe/TCP : SUCCESS in 0h 46m 3s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":12},{"id":"a30dc3bdbc368b9814cb51d0aeb50129750f25e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-05 09:52:19.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_10_2026_09_46_16\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"79b0c60864516cc4e85394a07dd9111f36042b84","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-05 10:27:48.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d64d6826712a49a6a0a5745de4aad318 : SUCCESS in 48m 17s\n- cinder-storpool-tempest-iscsi-multipath build/fef14c42e6d044cb9a6b53ce43af743b : FAILURE in 1h 02m 54s","accounts_in_message":[],"_revision_number":12},{"id":"a3a71fb8638a86dd2819506f430b604e5b1d5f4f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-06 03:04:11.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5845/.","accounts_in_message":[],"_revision_number":12},{"id":"10d105a7f7700ecf15819f84306613bc6e90a06a","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 08:50:58.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"39396c81ca70aeeb2c7e8e7325e837ebb7e1edc4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-06 10:37:28.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/48b61e4100e34ac4bd936b73582e0401 : SUCCESS in 46m 11s\n- cinder-storpool-tempest-iscsi-multipath build/c7552acdba40430b96c59aa4c65f9594 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":12},{"id":"b75edd576b1752fd44c0f3bb770f836ab7d8e299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 10:37:37.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"4252899c74e11510718ce81e123762c9dbf2d035","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 10:48:32.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"47b455272ba035eabcb25a18ed4850f261605b90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 11:02:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdcf8dc96d04ac3abcd5106e71cce3f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/783ea113dbe24836b996e5b1eed9fef6 : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4fa665d664904289982cb410307d56aa : SUCCESS in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4da460a5a707635a74fd2b122b7b86d7b2615670","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-06 11:12:35.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3c2a38b3c66e47d18badfab89d4f1e45/","accounts_in_message":[],"_revision_number":13},{"id":"b13aa723ae0e059e89c3ac0693fc3fe9578e5ff6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-06 11:29:24.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3c2a38b3c66e47d18badfab89d4f1e45\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb455888f9ed45bb84851f4d52079478 : SUCCESS in 37m 52s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/447840c2012d483e868ad05217547ae7 : SUCCESS in 34m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8b425de74234eb3bfbdf5ed1494852f : SUCCESS in 42m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4c85a36fd1914a48a1dc61502d5e4fc5 : SUCCESS in 28m 28s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa6b2550b4df4ba28e7d82ff2b14f13f : FAILURE in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"15084e7ec553922c1a0fe013b26f4e8c301d0fc9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-06 11:41:02.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I45d32892e7712cb3bb4e52cf4547691505b01d01 .","accounts_in_message":[],"_revision_number":13},{"id":"08c9213df3fb3a6d14991eaeb17cef6c4004daa2","tag":"autogenerated:zuul:check","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2026-10-06 12:14:37.000000000","message":"Patch Set 13:\n\nBuild failed (LINBIT LINSTOR CI). Comment \"recheck\" or \"run-linstor\"\nto run again.\n\nhttps://zuul.linbit.com/t/linbit/buildset/08e1dcb735954fc8a3f82bc4c80478a6\n\n- linstor-cinder-drbd https://zuul.linbit.com/t/linbit/build/996eed1533084d4781b115ee1607ae54 : FAILURE in 1h 32m 50s\n\nLINBIT LINSTOR CI tests the LINSTOR Cinder driver and does not vote.\nWiki: https://wiki.openstack.org/wiki/ThirdPartySystems/LINBIT_LINSTOR_CI\nContact: linbit_ci@linbit.com","accounts_in_message":[],"_revision_number":13},{"id":"bafdb4cfcf41e78755b76b6d4f6cf781acf22264","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-06 12:37:15.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5681/.","accounts_in_message":[],"_revision_number":13},{"id":"5aa79100ab95429c72cda6322f95d46dda2b947b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-06 12:37:57.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/fad98c6a82cf4cf08b4c088289d0d446 : SUCCESS in 50m 18s\n- cinder-storpool-tempest-iscsi-multipath build/cfcd7c71743e4568a3b2b615dd32531d : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":13},{"id":"5aa3c899b027c6edb4c0d5d75bc2577334db94ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 13:19:28.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2274259d075f443aa64dfc7700b031cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c081f8f5dece4eee8a7e72648cf2c21c : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/862d8c39574e43cab6739ae55305e744 : SUCCESS in 13m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6aa8416173bb4d35b1b770c33888fa12 : SUCCESS in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23039c936dc8484abe7ac32da620a6e1 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b0725f4955ca4427b97d1b08159fb6db : SUCCESS in 1h 12m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e4038e27848486ea76b444b37ab0546 : SUCCESS in 1h 08m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc938adffe1e4a5295c55979a3e684b0 : SUCCESS in 1h 34m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ac736fe1674accbd28379177ecd353 : SUCCESS in 1h 00m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ceacf199ea4168a9017d0682376c56 : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eedf2709f2874088b8b6786718436ffd : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2468df5dd66744c9bbc7beb077292b2e : SUCCESS in 5m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa32ceab8c6c42e7bcfee8abdefa1edf : SUCCESS in 8m 18s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1f92ac20421f475cb118ddd126fbdfcd : FAILURE in 10m 54s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6ba750bc92954369a05bbe625f6ae16a : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/27740529726643db989f91006075fa2b : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5fcb5b1cca864f1d9788a20f4f1518c3 : SUCCESS in 5m 46s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a3639a724761456ba51d61fb6be0c6bf : FAILURE in 6m 12s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/dfa3cf164aa84e42b53bb113c3a03851 : SUCCESS in 8m 55s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad9cd390198f4900ab46eac59cddf58d : SUCCESS in 51m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89953bbf5ed746af9c34aec0a50ab165 : FAILURE in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eaf388daebfe44ae9887892f79f11cbc : SUCCESS in 1h 52m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/37255416baae49429db6e8fed3e585fa : FAILURE in 2h 05m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25c67fe9537c439ebb954730168d7a14 : SUCCESS in 1h 52m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9fa961b233e74fd88b63290d1bb38aae : FAILURE in 2h 34m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/223f6c176dc045b3b76f7617bb0cca65 : SUCCESS in 1h 45m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e474f7e2f35a4e5180a669113246907c : SUCCESS in 48m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/19218243deed4d4d9b04612a48038c67 : SUCCESS in 1h 02m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2a642c17bce14fec82562802e2c98820 : SUCCESS in 1h 59m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/143a8762fec2435790a1b15f0284681c : SUCCESS in 37m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f976b4ff0b6e475bba91f6b14ead2d37 : SUCCESS in 1h 27m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81cabf13a4f74df6be77fd0f6b5ab57c : SUCCESS in 1h 11m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"18a4ecef069fdc0e1befa78675678d3846170ebc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-06 13:32:35.000000000","message":"Patch Set 13:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 53s\n- FC       : SUCCESS in 0h 48m 46s\n- NVMe/TCP : SUCCESS in 0h 48m 54s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973592","accounts_in_message":[],"_revision_number":13},{"id":"eec8d26c5f247928fa46174a694a4fee35afbc81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-06 13:36:14.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_10_2026_13_30_05\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"88bcf24133f454018b598205d5bb1aa76c5eef85","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-10-06 14:17:38.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1dcc813852f276d5f5b759991a5e4a4e87db1aa6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-07 14:42:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5860/.","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"acbe972d5a8e2e648dce55cf3a5376767e97a6dd","revisions":{"58e4987b543e6b6a954a7c8537b5f5ae33f58a3c":{"kind":"REWORK","_number":1,"created":"2026-01-16 10:04:44.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/1"}}},"commit":{"parents":[{"commit":"bfbeb25cb0e44a14d00610deec5c86632a5e1251","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bfbeb25cb0e44a14d00610deec5c86632a5e1251"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-15 14:42:51.000000000","tz":120},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58e4987b543e6b6a954a7c8537b5f5ae33f58a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58e4987b543e6b6a954a7c8537b5f5ae33f58a3c"}]},"branch":"refs/heads/master"},"03e4b5c299a4381ff53ee9b7d3b0fb88275ef093":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-02 07:57:13.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/2"}}},"commit":{"parents":[{"commit":"c39aef4c072e1d9018251b8f4a81f069465ab853","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c39aef4c072e1d9018251b8f4a81f069465ab853"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-04-01 14:53:12.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e4b5c299a4381ff53ee9b7d3b0fb88275ef093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e4b5c299a4381ff53ee9b7d3b0fb88275ef093"}]},"branch":"refs/heads/master"},"9351c9aed853191fcd2edb99aa1523c897d30fcc":{"kind":"REWORK","_number":3,"created":"2026-09-30 14:22:15.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/3"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-09-30 14:20:09.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9351c9aed853191fcd2edb99aa1523c897d30fcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9351c9aed853191fcd2edb99aa1523c897d30fcc"}]},"branch":"refs/heads/master"},"2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa":{"kind":"REWORK","_number":4,"created":"2026-10-01 08:54:08.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/4"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 08:08:15.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that is not used anymore\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa"}]},"branch":"refs/heads/master"},"e6acd8df4ebd32b018237583e0c8ced68fb8a82e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-10-01 08:59:33.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/5"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 08:57:04.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been depracated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6acd8df4ebd32b018237583e0c8ced68fb8a82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6acd8df4ebd32b018237583e0c8ced68fb8a82e"}]},"branch":"refs/heads/master"},"1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-10-01 09:00:25.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/6"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 09:00:15.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10"}]},"branch":"refs/heads/master"},"9d120098274a470ef25bf7a67b06dcc179d6e49f":{"kind":"REWORK","_number":7,"created":"2026-10-02 07:50:07.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/7"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 07:36:16.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d120098274a470ef25bf7a67b06dcc179d6e49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d120098274a470ef25bf7a67b06dcc179d6e49f"}]},"branch":"refs/heads/master"},"2b576744798b6e9df42ec177c1d4967ff7d43c34":{"kind":"REWORK","_number":8,"created":"2026-10-02 08:28:47.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/8"}}},"commit":{"parents":[{"commit":"d6f478985c614bce6b9b518e9241909119c861bc","subject":"Stop depending on Python packages storpool and storpool.spopenstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f478985c614bce6b9b518e9241909119c861bc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 08:28:40.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b576744798b6e9df42ec177c1d4967ff7d43c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b576744798b6e9df42ec177c1d4967ff7d43c34"}]},"branch":"refs/heads/master"},"56cd9e4c52a756bab98912da1754ac9a3464dff5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-10-02 12:03:25.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/9"}}},"commit":{"parents":[{"commit":"a7e8131c2fd820fbe5e4b04500dfff94a40b9688","subject":"StorPool: Fix a leftover dict unpacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7e8131c2fd820fbe5e4b04500dfff94a40b9688"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:03:14.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56cd9e4c52a756bab98912da1754ac9a3464dff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56cd9e4c52a756bab98912da1754ac9a3464dff5"}]},"branch":"refs/heads/master"},"8e9192e6d4ec283a05fb890c17276e997ec7a043":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-10-02 12:36:55.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/10"}}},"commit":{"parents":[{"commit":"be699684357e53827d34158cac1255f1b36b68d2","subject":"StorPool: Fix a leftover dict unpacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/be699684357e53827d34158cac1255f1b36b68d2"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:36:24.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e9192e6d4ec283a05fb890c17276e997ec7a043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e9192e6d4ec283a05fb890c17276e997ec7a043"}]},"branch":"refs/heads/master"},"0fdc1ea96665097d3c8408a6e3187526d8bd7b47":{"kind":"NO_CHANGE","_number":11,"created":"2026-10-02 12:41:38.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/11"}}},"commit":{"parents":[{"commit":"0beb19ac1af931d0f244853bf5b09c3850cb91ea","subject":"StorPool: Fix a leftover dict unpacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0beb19ac1af931d0f244853bf5b09c3850cb91ea"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:41:32.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fdc1ea96665097d3c8408a6e3187526d8bd7b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fdc1ea96665097d3c8408a6e3187526d8bd7b47"}]},"branch":"refs/heads/master"},"5f0fbce1942514a98a330a662dfb9a017445119d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-10-05 07:04:56.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/12"}}},"commit":{"parents":[{"commit":"b8039badf872aad6a4a2a194819e2d5453272dca","subject":"StorPool: Fix a leftover dict unpacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b8039badf872aad6a4a2a194819e2d5453272dca"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-05 06:54:30.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f0fbce1942514a98a330a662dfb9a017445119d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f0fbce1942514a98a330a662dfb9a017445119d"}]},"branch":"refs/heads/master"},"acbe972d5a8e2e648dce55cf3a5376767e97a6dd":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-10-06 10:37:37.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/92/973592/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/973592/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/973592/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/973592/13"}}},"commit":{"parents":[{"commit":"92d75f3fa29711ed9ddc898db59806c9787bc8c3","subject":"StorPool: Fix a leftover dict unpacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92d75f3fa29711ed9ddc898db59806c9787bc8c3"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-08 14:55:49.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-06 10:09:45.000000000","tz":180},"subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","message":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes\n\nThis is a legacy option that has been deprecated for a very long time\n\nChange-Id: I45d32892e7712cb3bb4e52cf4547691505b01d01\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acbe972d5a8e2e648dce55cf3a5376767e97a6dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acbe972d5a8e2e648dce55cf3a5376767e97a6dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
