)]}'
{"id":"openstack%2Fcinder~973593","triplet_id":"openstack%2Fcinder~master~I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa","project":"openstack/cinder","branch":"master","attention_set":{"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-10-02 08:30:12.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2026-10-07 07:12:25.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_35429\u003e by clicking the attention icon","reason_account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"}}},"hashtags":[],"change_id":"I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa","subject":"StorPool: Use only baseOn when cloning a volume","status":"NEW","created":"2026-01-16 10:04:44.000000000","updated":"2026-10-07 07:12:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd15ca5c7b76b4f461d0f7505c4dc8980771c4b5","_number":973593,"virtual_id_number":973593,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-06 13:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 10:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-16 11:21:09.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-16 11:25:36.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-16 12:23:44.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-16 12:47:16.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-16 12:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 14:47:42.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-01-16 20:48:50.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-02 10:09:33.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-02 10:54:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-02 18:49:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-04-03 03:52:21.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-04-09 03:49:12.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-25 08:33:18.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-06-19 19:56:23.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-09-30 15:14:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-30 16:56:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-10-06 12:17:00.000000000","updated_by":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"reviewer":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"state":"CC"}],"messages":[{"id":"b25c5574a3ce3b57814ffe6530967ce9877886e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-01-16 10:04:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6423e5bc756987b7c218bdc09616efa82a69cfef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 10:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b90a812e25476ca7dfd36e1664997e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/114694c126de41ed83b72c35ff0ea3f2 : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3d1500a47f148048fcb046031b031cb : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/388be69a8d704c24b9aaa2e7fa3b3b98 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0a1e031c89a24815fc979c4a5861477a79ee1ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-16 11:21:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":1},{"id":"7d46fe6d24d92441cfeabe3f8966c667c8a56abc","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-16 11:25:36.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_241/index.html : FAILURE in 80m","accounts_in_message":[],"_revision_number":1},{"id":"e9de6677145a3c814a129c65d91f82137324a813","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-16 12:23:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/0309fc51e59447ada9d330c1a4c52ea3 : FAILURE in 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4682bc0596eb5775d7ebf37d1a535af6a99750c9","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-16 12:47:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d9a8a562ed3547d990aba4b4b6376034 : FAILURE in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"21449846099738c424f48566c5f9434ae36ee00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:53:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8ea383769774097b01cf5bf64d4cde7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae98ca0a1c0f4358ac123c1ad1eaa9af : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/950b8f84ada44bed8237266f17d62cee : SUCCESS in 11m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef27acb6759245aaa6e508038a989302 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e55e0a2fb14b1f885e9565a4f7b52c : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/89014c73d19941d48c4340018971c91e : SUCCESS in 58m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/490dd9aab4f5438aa40e7691c076fdfc : SUCCESS in 1h 06m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9491d28253764ec296bf347fb98ca767 : SUCCESS in 1h 32m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd570a9eb2d04728ad6ba9f31c9e1b95 : SUCCESS in 58m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cedf24153b794459869477aab6026ff6 : SUCCESS in 21m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0a9eb4e7dc484572a684c4d67e29adcb : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9c5b7126ad04df78eec80f439c22b98 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf9af32e4af54f88b5a5fbbbad85d832 : SUCCESS in 6m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3f75c6e8a8e84a2dbcfabad5afbe4df4 : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60925228475948b48e5ee9b27cfe6d60 : FAILURE in 38m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5724de7ad64492caef3f98e66895fd3 : SUCCESS in 6m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5947cbe6ff86411c9617337592d9e96d : FAILURE in 18m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d096a758b0b24dc094498d21fc2c4854 : FAILURE in 2h 41m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9c74570e6d3f479584d1db60388a9627 : SUCCESS in 1h 34m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/80313b628aca482286e7984d8a1ddbd2 : SUCCESS in 2h 11m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0217dd9174b64120a881d9eba3e36c66 : SUCCESS in 37m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06ea87427b3c4a57bc5c51710032cd7f : SUCCESS in 1h 05m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/24a55080c21643b9b7bf3de100442cf9 : FAILURE in 1h 46m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/625c2273c87f4693a33d267d4deb5e66 : SUCCESS in 37m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/01baa91cfe414939a58e90e5095da23c : SUCCESS in 1h 36m 21s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/703a9a81a43349a795fe62ebae31c907 : SUCCESS in 1h 55m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ad8747c8c48468eaa5402fe44e3e0bb : SUCCESS in 56m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9158208f30845c3909a4a16cdb01299 : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d12707a20963719a54825861e5680069c74ed316","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-16 14:47:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/973593 : FAILURE in 1h 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"eec1ab3e9a9c1a46d3af2cbf6ca661a543c55111","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-16 20:48:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4ace7078a018422a8f7be87d5c5443d5 : FAILURE in 1h 11m 41s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14496de1a28a424885ef1b2595af26ee : RETRY_LIMIT in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ce7522039711d8d094c5b9fab7633a8ca1aaafd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-04-02 07:57:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"02e3aa25b1ed2802eca8a8e9d58477462f946ad9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 08:19:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7e6f87b09e7445f88d8885f515df8f9c : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/48621429960f4112a61c3460c70547c0 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":2},{"id":"b60a6a2a874755a34709bb71311a00ad9b7ac3e6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 08:38:28.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":2},{"id":"7091d33ebc2c6d689b40222edc29cf0858da2914","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 09:39:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f963f4c2c2430da92161dcbc4f2b97\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29935b8e9dbc47ed9348f5b34f70dd56 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a98884b44524fb88f565dba8afe9420 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e11abab497d4cc780293feeed8acc05 : SUCCESS in 17m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61e59028a3ded83a6dded3d67fbdac4f7987bdbb","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-02 10:09:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/b8074dbefc46481fb9bcf09ea700b457 : FAILURE in 33m 27s","accounts_in_message":[],"_revision_number":2},{"id":"06218c390fb22b9acbef75ae78690825e6afb4de","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-02 10:14:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/221e7239b551447089f5ad2d921e7188 : FAILURE in 41m 33s","accounts_in_message":[],"_revision_number":2},{"id":"efd057477c19dca57b0b11faa8e51a544ed2d143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:25:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c88126bbb4854a1eaf1b188cb9ce9b93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752c4ed7dc7641f6aba5eb25fb986523 : SUCCESS in 8m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a80a4e169724e6688e71476500187cb : SUCCESS in 10m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd43b756fa214bd09fb729aa981d8b6d : SUCCESS in 8m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b8728f7cb46d408dae87a4e549d7559d : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc069e2d1f1440f6b60edcfcd7573f15 : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e42b507a5ad949a58f6a42aaae2ea940 : SUCCESS in 54m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ecb422a16234032a08c6bae60506c0e : SUCCESS in 1h 02m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c227de104387403783ce37d4985486e8 : SUCCESS in 1h 34m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0931a6a4bfea4f3c977c6bc5282dd71e : SUCCESS in 1h 13m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ec64bd78bf6f4309af638fe825364b2c : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0c5e06b8ac834a919aaf769692f4c852 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/000908b22c8144b398fd4bfaad915a6c : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc6d1a7f17d2469ba7b3441dd9215353 : SUCCESS in 6m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/17e3993ab10443df96af2b6c9abd8089 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b8095521467411ab0b6d84b9826bca1 : FAILURE in 42m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14a5ed848a5a4475979b192953e2cb9e : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a560d6f6617405597354553fd932555 : FAILURE in 1h 53m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3726822b46bb4df28424e45ec2143401 : FAILURE in 2h 18m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3f13c630e2e46e9b3ff8e10458d5e03 : SUCCESS in 1h 54m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/00dc1cf5cae041989a677420b8a6ba89 : SUCCESS in 2h 06m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ddbb18578cd14fbf8ce0fb94648b5d36 : SUCCESS in 37m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd3cffc1092b4872886f93954949ea11 : SUCCESS in 1h 16m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f3c5cf61d3a64be5b5817fd86a322163 : SUCCESS in 1h 40m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dfabd14d6a0e474380d697fa1dd9dcae : SUCCESS in 41m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/74eb6d9b52c240f78bc5b70198c232a0 : SUCCESS in 1h 53m 23s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9ae393a112b247e3a2ba88064df0e52c : SUCCESS in 2h 16m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9746349f3039488f8010b3b735612346 : SUCCESS in 1h 49m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82a572f16aea471081cdde1eff872cc5 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":2},{"id":"93d9f0febb44f42bc19d5782244adf5cdf1188c1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-02 10:44:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/973593 : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"89d472e7332c71552c7f1e36c815096c1311362d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-02 10:54:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/426adf5442d8481f95ee6d51def2ac5d : FAILURE in 41m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7fd87d1eb9b70ec55b830e08c2f41599d926be64","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 18:49:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3683/.","accounts_in_message":[],"_revision_number":2},{"id":"67e718666f0da866c1135ca9240138dc18e412fe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-04-03 03:52:21.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-973593-2/FC/ : FAILURE in 19h 54m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-973593-2/iSCSI/ : SUCCESS in 1h 42m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"eaf67c3c3795cb7a1217c26e2c3d70e3a7867303","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-05 12:22:37.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3507/.","accounts_in_message":[],"_revision_number":2},{"id":"ebd3e9adb11c866b5b2e076bf4be2ebea71de54c","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-09 03:49:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3173ddd847714aed3a02bba4b1eb0f412985950e","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-25 08:33:18.000000000","message":"Patch Set 2:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/973593 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"69d0f3fe5f7f6253a2b19c2ff958200b5f58b6de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:23.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"7d6bba5f9c9e33d100e0ec39ac090ce8aeaf1817","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-09-25 07:59:10.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"e83de2b1a467922298f5ac930c74712a3e76d258","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-09-30 14:22:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"40100e4fd1d25888a021c43b2d5020e362716749","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 15:04:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cacc3fd255d478fb7ff70eada5c7a18\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18ac5a142ea541b7a3831cda62880b12 : SUCCESS in 16m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/35ae6e4a9ce54913b2d79d2ca60cd7e4 : SUCCESS in 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd3f8a28023d24dc4f65bf0717f198c53a1e69bd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 15:14:15.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/6d4ed75b4d374ace8df296cd4bce2cfc/","accounts_in_message":[],"_revision_number":3},{"id":"d115fb2cd01dc5dd6ac467ff26e653bc06f901e2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-30 15:25:58.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":3},{"id":"f9b737f639ea847e603bd4e7611541844bb63c3c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-30 15:56:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6d4ed75b4d374ace8df296cd4bce2cfc\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4845260251e44088033ebb9972fa08c : SUCCESS in 44m 06s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b253f834d5e4dfcb835278c65acaaac : SUCCESS in 34m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/642f466384d942e4a45c92bc88ce5809 : SUCCESS in 43m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/58dfa9de7a2b4fbe8083a5d67e8c1a1f : SUCCESS in 32m 14s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d913a3be5ba240db967bf230f84791db : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de4adaf4c92b8fb1c9f4384333a4765828a5c7d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-30 16:06:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ef4ba1f4e1784500858f22268be7c6cd : SUCCESS in 42m 16s\n- cinder-storpool-tempest-iscsi-multipath build/6447550703584b29b8ee69be6fd30ccc : RETRY_LIMIT in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"23b622f13992cae82e763aa0f03f34c58cf63449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-30 16:56:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d16ba6142641feb45f8feac7226275\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963c6c1ac94a45d88c1acfc44ad173ad : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/758e7775e5794c72b7ceb2d75de1f8e0 : SUCCESS in 11m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2310754881c4d02a4ff0a68b439f3a5 : SUCCESS in 15m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d75c4e53ba48ce855c40d005083b15 : SUCCESS in 19m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/1dd6c6969fc9429984b79dd1fb118875 : SUCCESS in 40m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81bd6e91a5094cc69ce340275840b060 : SUCCESS in 57m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c9d9f1a1c1649f6a2b872548effcb79 : SUCCESS in 1h 25m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb202496170f4d88985b6c3f79e91a1b : SUCCESS in 1h 00m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ced55982330041abbbbf9fd8f7f6914b : SUCCESS in 7m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d372be1427d84a3e9151a11cfe87fe83 : SUCCESS in 24m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f5624540abe4da1b4965783a425c66c : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af85a535964d45e08ad7c88101010ef8 : SUCCESS in 7m 14s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c7835618dc4a40d5bb5866d0f447faa8 : FAILURE in 10m 41s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d4090ce074e34c128f15766fe0e6601f : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/718f4b92106a423aadcf331237c5de9a : SUCCESS in 6m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/48c37ebe04204abf8cb982a001712356 : SUCCESS in 6m 33s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3cc6fdda8f434d4d9c9e0c12d6395107 : FAILURE in 4m 49s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/870969c10da5422e9fda0f39d4856402 : SUCCESS in 7m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05768f3b9538436690924d369453fa10 : SUCCESS in 23m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/79d0f9cc63bc43bdb58a8bb51edb552a : FAILURE in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c00e04f147c4c309b654bfa861d7940 : SUCCESS in 1h 22m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1688d4ed75814d77a08b65414af14d4d : FAILURE in 2h 22m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b67ee95254b94fbda91cfa84020c0918 : SUCCESS in 1h 42m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7a7e51a4bcda4d369393b4d22ba55859 : FAILURE in 2h 05m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/54410c9a41ae4ce6a0186a4b30cc034d : SUCCESS in 1h 49m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ca026e00a6bf4703b8f8fb044e4615b4 : SUCCESS in 36m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4654f35dfcf4a44af92d627707174bd : SUCCESS in 1h 14m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e240ae7243dd411b9c1eef5e841a6e4e : SUCCESS in 1h 13m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/855488cb78e8491eb9b18a7458b7859f : SUCCESS in 43m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/66c5406d03324c72b8a06d8491fc2694 : SUCCESS in 1h 37m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6945c7aea1c347d8b60149e2daf479da : SUCCESS in 52m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"dc8fe62ecd677eb12d84e3805ca5057ef704a071","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-30 17:06:32.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_09_2026_17_00_31\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"fbe1322c03501614fe81a55324929a3d183b5e54","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-30 17:09:32.000000000","message":"Patch Set 3:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 25s\n- FC       : SUCCESS in 0h 47m 32s\n- NVMe/TCP : SUCCESS in 0h 48m 0s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":3},{"id":"188935adff9c28f60f1d53c55cb57875a79e65c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:54:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f5894a368730ed458dfc16cdd00685a1a19e2c8d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 08:59:33.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"13dd017e992cb448396bacae948ca190467275e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-01 09:00:25.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"29d10be029d81979885caaff92386e8f1a6efff2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 09:06:55.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5776/.","accounts_in_message":[],"_revision_number":3},{"id":"b27289c358e3df7fa731ad09cb8c1e7ac9aed4f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 09:17:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e07343fe83649b4ba096dbbfa211ec8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/155834a2786747c69d31b2475308e9a3 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ef201329bf76413282db4f7cdc5b3c40 : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8dad12dad0beec061738630e8a51333b9be502cc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-01 09:37:38.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/05d76eda413a4db7a27caa55c2cbd576/","accounts_in_message":[],"_revision_number":6},{"id":"bef16828dca88bd7583cec5150d03e4991a83fc8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-01 09:48:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/05d76eda413a4db7a27caa55c2cbd576\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dab558e20d8543b1ac5693a980c43428 : SUCCESS in 38m 32s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cd6a607fcbb4015973fb6b0580b4ee9 : SUCCESS in 34m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e40ebc6a02fd458391451d6cdb111bd4 : SUCCESS in 44m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e43e70483d8042aca86d54302df151db : SUCCESS in 30m 56s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d1554b7230024949878dd0ad64d17097 : FAILURE in 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf77b911800b1938034ec84f7d9b44e0ce0e8aa8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 09:57:42.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":4},{"id":"115fa117f55117e6ddba8d7754fa28ee85141a2a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-01 09:58:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/429b5b1456294a4b9be383e946e5c46e : SUCCESS in 48m 17s\n- cinder-storpool-tempest-iscsi-multipath build/17e0a0c193d449b682ccdb03a8c255cd : RETRY_LIMIT in 5m 05s","accounts_in_message":[],"_revision_number":6},{"id":"00d1c3f006b8d29d011fb4aaf901d66a57655249","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 10:15:42.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":5},{"id":"39a939bcd1ebbe5b7a6f82b512d6472158009fd8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-01 10:46:14.000000000","message":"Patch Set 6:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 55s\n- FC       : SUCCESS in 0h 46m 33s\n- NVMe/TCP : SUCCESS in 0h 45m 9s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":6},{"id":"e5eae73dc1198fe91eed9bc14bda0369d5be9ce4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-01 10:59:22.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":6},{"id":"4c8c56f0dc3e54ea781100e8573c5da3b26683b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:16:52.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5599/.","accounts_in_message":[],"_revision_number":3},{"id":"fc62652f57e2b560463b061497945260e5fcd949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 11:19:31.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e6de2824e14ba6ab69b96f2ed765a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c2c116f52d4a81a7738d4920510574 : FAILURE in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89a639d9b0884e71ae5b68ea4cc88701 : SUCCESS in 11m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3fbd3552ccb0402ca4860a6c9252d726 : SUCCESS in 15m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf513f26f9b42d5bbf045e964ebd192 : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/ab3d9d457a1048a3a48b0a62147b9598 : SUCCESS in 1h 08m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5fb64c3d4ab2462583ff038da9fd60a2 : SUCCESS in 1h 05m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbbcec081ad045abb0518b63dcf40d12 : SUCCESS in 1h 31m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfddc96dee0a49afa5f9b9f468ee3541 : SUCCESS in 1h 01m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e44a274de7694ffb9e7ecabbfa21b5de : SUCCESS in 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7ac831ebb544ce8a7cc06b1c516e744 : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e703b7ee77b42da9d755b88b4b10b18 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3cf6a0c8b1de4315ac15d95419221966 : SUCCESS in 4m 15s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/3a3aab8c63f846ca84d8517090a7bb07 : SUCCESS in 13m 07s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/427ec66e9ccd4af9847bd975a4286a2b : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7df149f5f22b4f52b9340b59af52452c : SUCCESS in 9m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1d28541a398c495c94e6da872c20e341 : SUCCESS in 6m 16s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9514981705be48dba806167b60749a78 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8dcb1c95d0ea4b9390680c1a1e2a7cae : SUCCESS in 9m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e1e520005314e4b9cd01e6b4aff038b : SUCCESS in 37m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b83a043fe74e410fa00b6a7942e10946 : FAILURE in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b715be75692b4ddc958ec013b1e88ce5 : SUCCESS in 1h 30m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/735deba865224353ab2bcbc1e96e883b : FAILURE in 2h 05m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0455734612da437f94e0a51c1a9bcb41 : SUCCESS in 1h 59m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/eddd43cc79f84d3ea3e1dc22f7d0b71a : FAILURE in 2h 17m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9d8fb80fde384e96bc74091f4f7c9a5c : SUCCESS in 1h 58m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad0fa9c5f7904073b51402d5cd245b90 : SUCCESS in 39m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3fdf8d13d74144ecb8e4749ce386de39 : SUCCESS in 1h 22m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/998234174e044d0b965d7622d6de2e6b : FAILURE in 1h 43m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c6ebcfe64a84221b474abbab64c70d9 : SUCCESS in 38m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c0e98bd1cf9442a082330d1304797ce0 : SUCCESS in 2h 02m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75f89ab6239f44d3b7d4a32173b7cf17 : SUCCESS in 51m 38s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"e5ffd8599284cc4fda6409af3d7e5198556c862f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:19:33.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5612/.","accounts_in_message":[],"_revision_number":4},{"id":"9ed2da2b8afa4aebc808a625bda5a433bbaf05de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 11:20:23.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5617/.","accounts_in_message":[],"_revision_number":5},{"id":"d0e2b07afca64351a7dbc325beaff42a717520f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 12:06:26.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5622/.","accounts_in_message":[],"_revision_number":6},{"id":"3f6dea5eef22a036a0ae92ac30ae4979f18e3f18","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 12:32:32.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5623/.","accounts_in_message":[],"_revision_number":6},{"id":"bdf8d337cd84b5f3c7ceb6fdaf06670aff729c1a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 13:49:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5624/.","accounts_in_message":[],"_revision_number":6},{"id":"5575a42caeefb1a14bb4875f2942e1e095376dc4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-01 19:47:35.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5789/.","accounts_in_message":[],"_revision_number":4},{"id":"3bd09fe11d8e9a7330df7bf5e30f62dacf69fc73","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 05:39:18.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5794/.","accounts_in_message":[],"_revision_number":5},{"id":"095edd5f0599e3fd65ab79d85ab893c7fc01e7bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 07:50:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"225d19b501e724c86488c24b0befd32447f09e59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:06:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4acf36957ca4198b6354439b48e933a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc28e11516df4e63b1adfd0a9275f252 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a8344536e52469886d79b45e64438c5 : SUCCESS in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2cd71c05e2345d754fe0ccd8b461840befde1585","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 08:25:50.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b0330a0e4f684a8a91ef0896f0f167c5/","accounts_in_message":[],"_revision_number":7},{"id":"bcf7aafc01c51eb987297fbc2269e4afbc6fe841","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 08:28:47.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"232108e3497957b08b24d57706ac44b23af659cf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 08:30:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"5e7971dd80843ddcbc3091c24c90ac2cf474248f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 08:48:02.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":7},{"id":"b01477f1fb99f04b758e8ebbf60a0a4f59343631","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 08:52:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/233e6fe00924493b8209946751d44795\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe484cda144f4d8e95c60c059bea97a1 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c4ea6f0d6554ed889bd56a48647482c : SUCCESS in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bf1e5661db0bd3c96f018f48ab35d28c64866d12","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 08:58:02.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d60e1a1e665f4b7aa39990f94b29354f/","accounts_in_message":[],"_revision_number":8},{"id":"a507ee689524b767e92f424c41a3156da705bffe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 09:15:46.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d60e1a1e665f4b7aa39990f94b29354f\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85a417671462471282594ef7bee5c00b : SUCCESS in 42m 24s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b873742762b548b79bf2bc7343e157a5 : SUCCESS in 32m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6f04836420a4e119a27bd3312833356 : SUCCESS in 40m 49s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fab8c35b19a546f2942151e108bfebac : SUCCESS in 31m 42s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb8d20ba26644b499042d03dd87b4172 : FAILURE in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e9f43b8efa2fcc3fc3159690209789196f719ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 09:28:23.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":8},{"id":"e90b40f9993f53398acb5d859f621dc6366c147f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 09:31:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5ae6ccada30c4f7f890d78f88a11a971 : SUCCESS in 54m 29s\n- cinder-storpool-tempest-iscsi-multipath build/db48b46f6d51439fbe4994a0d971aac5 : RETRY_LIMIT in 5m 44s","accounts_in_message":[],"_revision_number":8},{"id":"9f821d1f95cd911983096257a4bb0debf7d48ee0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 09:43:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5627/.","accounts_in_message":[],"_revision_number":7},{"id":"4c240f658825629d071939dc891265e4e592a7f3","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 10:08:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7cafdb06c57768eecaeeec960bf4545a21850746","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-02 10:29:00.000000000","message":"Patch Set 8:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 17s\n- FC       : SUCCESS in 0h 47m 37s\n- NVMe/TCP : SUCCESS in 0h 47m 4s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":8},{"id":"1f6da8e4598df167fc822027ced355cbd0236032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 11:05:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b53fc5be2a40dbb3bf5c844dff50f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dc441baa4f43438ddd5cb5cac7722d : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6b193b3648f4a3a9bcd436180e79758 : SUCCESS in 10m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96cc6a870a9b44eca6611de61faba183 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e134203df46c4f3a98e5bb4995c0ef13 : SUCCESS in 14m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/255b522841234c1085a7bca6af145418 : SUCCESS in 29m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af56b7d832b64e10af3d46ddd5341f28 : SUCCESS in 56m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7c82816643046679f8d8263c344b102 : SUCCESS in 1h 26m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72b803aa4e5346f8861040c36d20f601 : SUCCESS in 32m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87ea871fb96e491981c83655dde19326 : SUCCESS in 9m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5eaab17be6374b5fb0e19a704dde6956 : SUCCESS in 22m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a69dc6c37ac04a18af5f696340f253b5 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/afa854b9ef6a4251a68162ceab5540ce : SUCCESS in 6m 48s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a60754895d284db08c4c212c981f28f1 : SUCCESS in 9m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7a821f33385147c5b9a22bab41c1d64d : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b59403f6f7fb44d7b0dd6e40b04d665f : SUCCESS in 7m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1de07db483d44d34a3085cb37afac28c : SUCCESS in 6m 33s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a640b2e6a3fb4088891a8459c077cf61 : FAILURE in 6m 00s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5b3663fadb15464fabc023fd37c03532 : SUCCESS in 6m 37s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b19a64ea89d64ea8bf2760c5834b0581 : SUCCESS in 37m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4523666dd1964420828b9abe99c99eee : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55e5ef4f587f4dceaea961cb058bda24 : SUCCESS in 1h 39m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ad704afee83478f9244c1b320b8887b : FAILURE in 2h 03m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af7175a724fc4c5aa5a85a81f24dd389 : SUCCESS in 1h 33m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d4b9a76afaed43139bcb83c6300fcfa2 : FAILURE in 2h 29m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ffdda747d3f6481c9dabe99ba5840366 : FAILURE in 1h 51m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/154d0c399133498990c955d12b268e7f : SUCCESS in 34m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/71abca5f8ed24ba4bcd8d9dffb896336 : SUCCESS in 1h 12m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61c5b8d4f76043678054528b8d0837ec : SUCCESS in 1h 40m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/735557407c1648e6b34660f8cb1b1669 : SUCCESS in 35m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35671cc6c8494b92a2eac04f6f50636c : SUCCESS in 1h 22m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ecca85d776d4335930f9200bdd9ddff : SUCCESS in 57m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"0a151a712d73c77b753e62320b17f8078556d2e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-02 11:14:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_10_2026_11_09_08\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"8a347fc12a3552b642ba053c30b8d46ad846bcd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:03:25.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b4e2dab1c0f838902abab6e635fc9e609c1100d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 12:36:34.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/117e0a0346614d41908b9fc7b20888dd/","accounts_in_message":[],"_revision_number":9},{"id":"5390c55cc889202d6ec7777928434e95efd59f96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:36:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"bbcebe1e11e5e7c38fb44ed714dc8d9b72c4ed04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-02 12:41:38.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent tree, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"a25adc4ebc143e4f51baecb3e0d6f267098d9763","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 12:49:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0ae60255614e339bc34570b2647a0f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f908a7fea3f2442e8edb6f14039d68e0 : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/42d5aa80657b4eaeaab2e724cc874736 : SUCCESS in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"eac9e9a454fde9510d7c9efb20a13c56cab54c42","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 13:05:28.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":9},{"id":"b9198dedb3eb5c702ae323aebe8ce85fc95a7b00","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 13:29:00.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3e710abd52c0441aab8ddee971cb5112/","accounts_in_message":[],"_revision_number":11},{"id":"4d83894bad64decefd19c2fa1d03fbc281df67d0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 14:02:18.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":10},{"id":"a0e7af8f629fba6ee212e528268f15b0803614c6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 14:14:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3e710abd52c0441aab8ddee971cb5112\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1bb7c39146104860b4a26972833613cd : SUCCESS in 34m 55s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5da03f5388dc47dca1884cc23deb3a5a : SUCCESS in 30m 39s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6cd7c3274b0d4923a7d9e08b6f5583e5 : SUCCESS in 42m 35s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23ec29d54ed2478a99a7d617e0f4409c : SUCCESS in 29m 17s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd410985d0df48f7a13ab3bbb6395e4f : FAILURE in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"49fe01ec5dfc1888d928ca2d63086edc73eacc09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 14:27:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/464e7e6152e740ce8cc505ea659410b4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52d6f0a6f00348a7bdcfe58a7c249a4f : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e82e46148f048e2a08468bdab08224a : FAILURE in 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2fa1cefbce0911a11a3df027d820f23d8e099675","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-02 14:33:50.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":11},{"id":"dd13faada983f48923e895bac5d4ceeb5497d79d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-02 15:29:20.000000000","message":"Patch Set 11:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 1s\n- FC       : SUCCESS in 0h 47m 53s\n- NVMe/TCP : SUCCESS in 0h 47m 41s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":11},{"id":"b01ff76942ce216ff725b09dcc93d381f10fc957","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 15:37:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5633/.","accounts_in_message":[],"_revision_number":8},{"id":"b885be7feb3236d47a8a05c98b5435f6708b2ee7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 16:11:34.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5799/.","accounts_in_message":[],"_revision_number":6},{"id":"ca8dd164f0cc153a614ae97a34faa85dbc9b5dab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 16:34:06.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/420c21aca9894f2584f1f213eb606769 : SUCCESS in 46m 53s\n- cinder-storpool-tempest-iscsi-multipath build/6447a2f357e24ef682e18c7dd8b4ae77 : FAILURE in 1h 12m 22s","accounts_in_message":[],"_revision_number":11},{"id":"d6573d619b3745a6338e082280db89376665843b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 17:22:54.000000000","message":"Patch Set 11: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40fb5cc9165d434393e1fb32d7ccd59f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fafef3da4074baea6065f128031b181 : FAILURE in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ccaa04ffe5d439496adff74e4b558c0 : FAILURE in 11m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78e31f8e4a1444b48be5d0ac81d8fa56 : SUCCESS in 16m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7ecad18f6142018d3586da4dbc16c7 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/1e326d262c1a44a29eaa75dd3ec30a07 : SUCCESS in 32m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/436b4fb659034c96ba9bde74debeafa3 : SUCCESS in 1h 19m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4acb12a988fe4a159342a0d516f533c6 : SUCCESS in 1h 19m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/445bbc6da1194ced8b4ed9ae4d33debd : SUCCESS in 36m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b0fe6a08a5c443a879ef0216fcc0676 : SUCCESS in 10m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f7508dd9f9c64bd8a0dca620de6ea3c0 : SUCCESS in 22m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/00a25b10cb934e6583e93f4ed8cf987a : SUCCESS in 7m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2221627e87cf446ca764cfbfecae9d28 : SUCCESS in 6m 10s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1785d35e5e6c431b8650976b99325317 : SUCCESS in 12m 06s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/687ad566e04a43cf9805e402d897c7e5 : FAILURE in 10m 27s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/09c30c1f0fd346888657733ecba573ff : SUCCESS in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/49a7b6d59b1245da982e5e1a8e344df7 : SUCCESS in 6m 56s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9bc480040dd343e38861d3c96aa9663e : FAILURE in 6m 25s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6fe0b9b2e9294434adf3750b681f90e8 : SUCCESS in 9m 37s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acc340af6c7d437ebe29ce3391a9f1e5 : SUCCESS in 39m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bcb079e988a4e049267dbb12edb0526 : FAILURE in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d3662b70d714abca184e4b9fc5fd803 : SUCCESS in 1h 42m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5016388efa9a433c883e779f81e02e47 : FAILURE in 2h 18m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d6a79049ba84c9a85aec847f6b4ca45 : SUCCESS in 2h 01m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8abaad5c356d4b5db07b5f477068ddfb : FAILURE in 2h 45m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/78ea26e024a54f27bef574255fd15cf3 : SUCCESS in 1h 50m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c1a1ad28d3ed44259467b88205de292d : SUCCESS in 38m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d261d7fe1584b0c93d3c8cc656c10f1 : SUCCESS in 1h 15m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f3a1ab700874500873bb039c80c5cbb : SUCCESS in 1h 42m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3907202436374af19efa1c554303f965 : SUCCESS in 45m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f89e297f314b4033bd3c34c745046209 : SUCCESS in 1h 34m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e7e2986f15443e18a85a241c37563fc : SUCCESS in 59m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"a316a783b49e64b186be593ecabddccc6ff5b304","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-02 20:00:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5638/.","accounts_in_message":[],"_revision_number":9},{"id":"1a3c219f06fc0ceac08748ba3d5d764f9e2aa24f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 02:47:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5645/.","accounts_in_message":[],"_revision_number":10},{"id":"811ad368d771c65e7998aa8013aa9bcfcb8236e4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 06:24:29.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5806/.","accounts_in_message":[],"_revision_number":7},{"id":"c03bd15dc3612c1d440664fa21e55ccc35811c2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 07:47:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5650/.","accounts_in_message":[],"_revision_number":11},{"id":"0d8d629ce79dd567ae7fdd406e31a2261274320c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-03 17:35:29.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5812/.","accounts_in_message":[],"_revision_number":8},{"id":"7fb9d05109cdd385c741ba0ac6f7b846bea2ecb5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-04 02:13:52.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5817/.","accounts_in_message":[],"_revision_number":9},{"id":"00637b1f8c5a830ea0f4001d38374b1de13b208a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-04 14:03:34.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5824/.","accounts_in_message":[],"_revision_number":10},{"id":"e91bfbb6e9b0bc379c0ec81e67824b4df1b27829","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-05 00:08:22.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5829/.","accounts_in_message":[],"_revision_number":11},{"id":"933f3019c9297113f8ba13c82ad27898aa891800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-05 07:04:56.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"bb3357661e8a4cbdf5e3e3e1c65c9712b815a3c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 08:01:57.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/fa0b951a98e94c1aac1f459cfc501888/","accounts_in_message":[],"_revision_number":12},{"id":"bddf1c61daa5c9af6c9b43dec72eb1e448a62289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 08:08:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/768898a455ef4bdd879a008cfde55d42\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3854b11ebfc54be0934494ece3882af1 : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/078bcc83572e42af872b39080a8f9db3 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"390504a7c928feeb2f6a1ddc8bcc8e395a28921b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-05 08:08:57.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":12},{"id":"ea3e19ef773ad21d079a6eee4c4b7d43e03d8165","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 08:49:47.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fa0b951a98e94c1aac1f459cfc501888\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e56a5e76732046a59835c3eed17f537d : FAILURE in 9m 01s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/394828b0d3e94a60b1edac0e680c3fff : SUCCESS in 30m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61fdbe85689541c79cbb0cf7535352b7 : SUCCESS in 51m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc6a1326d25340db9b72bc4b0b0d99cf : SUCCESS in 30m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cf8e51ce74243e1940bacd289000c15 : FAILURE in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"55d47128211e6867659ea052003f7a2e5428b726","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-05 09:04:15.000000000","message":"Patch Set 12:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 1m 21s\n- FC       : FAILURE in 0h 1m 3s\n- NVMe/TCP : SUCCESS in 0h 46m 8s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":12},{"id":"bd8897c2402522f2c8a9b9ca7bdfece108208365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 09:44:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db312e93cd0c43e288c5e99712f03b44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e974cb158834c04a4e43ab4dc66bb2c : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a2b0176450f478eb4bdaf3b332938e7 : SUCCESS in 11m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d59e77fe3ed2436390b65e5cb4a8a531 : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a00ed235724543821b79a473a563a3 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/dc6bf60bfd3f483ab023497dfe5f721f : SUCCESS in 1h 01m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6842c96b60943b0813fc74f3e37dfde : SUCCESS in 48m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dcdaff904cef454fbeb3febedc1fef19 : SUCCESS in 1h 27m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b84d02a7a4304d82aae1c1429ac1e3a9 : SUCCESS in 1h 06m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18aa5f728043423e87a49767acddd5d1 : SUCCESS in 9m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/22d899ae536141abb5b1eeeaff27d5d1 : TIMED_OUT in 41m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e2263541f5db4a439adf7909d859baf8 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/90585f1c771941308ab9faf1fb54089e : SUCCESS in 6m 45s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/667135271cc941b78ff173372dda627d : SUCCESS in 12m 57s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/26a81359cedf45128b583849087980bc : FAILURE in 10m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f0ca6c90bfd4443287fc35ab01ed736b : SUCCESS in 6m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1a1fa6e73c7b457890e0ade041c4a177 : SUCCESS in 6m 16s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/01519aa0c166486a9686d190ff26f921 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d67626455bc043cfada38411c534e42d : SUCCESS in 6m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7fc4a00a7e934f139c1bc22000bbff90 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c945ff8806b84c1880321e1bf2f99a5f : FAILURE in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be3ebe2cedf7427dba6c53bbfc6bd53a : SUCCESS in 1h 36m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/305a4e710367494faada639512f7346f : FAILURE in 1h 16m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/534778e061bc41aab6cf5427564084cd : SUCCESS in 1h 43m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6a28fa77b0eb450695360e087302174e : FAILURE in 2h 33m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9dbe989c8f2347ab951f57696e7cba44 : SUCCESS in 1h 46m 08s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1074eb9252194e61b548bb100fe10241 : SUCCESS in 35m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91353d9508cf46c0a10a56331095ce8b : SUCCESS in 1h 16m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/67afcbc3d31b4d1582484dcddd25c683 : SUCCESS in 1h 26m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a72668d2d1e64cb69b7d06df3693e6bf : SUCCESS in 32m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ce098ba8a084bd7b25b403ef5f3b8dd : FAILURE in 1h 38m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0db3ff348aa41fda90176eb45a387dd : SUCCESS in 42m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"86551afad085fc7ec6e9a7e0a030f676efad77a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-05 11:01:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5668/.","accounts_in_message":[],"_revision_number":12},{"id":"793e10413a91a2e219bc498a6cb31306b37d1e7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-05 11:52:50.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest build/aedcf25ab4474b93abd68a33f07b35f7 : SUCCESS in 45m 54s\n- cinder-storpool-tempest-iscsi-multipath build/a5cfef7dca374570b35222461b4195de : FAILURE in 1h 23m 57s","accounts_in_message":[],"_revision_number":12},{"id":"176b25eea65799555bd412f23354b2cf4ea34366","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 10:00:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fe1b04c0fdb029bfe8e288d14e7de72f74ddf06a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 10:24:11.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301ce5f7fed74ec98fb23c3b9864579b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb852640bedb49b1aedd58092e33f3c1 : SUCCESS in 16m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/580a671103d849c1ad778df2d161e0ce : SUCCESS in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f5b194f6634f37f92526f45ccd350e32bb77e14f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 10:37:37.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5c8869197494acd49683cb3975c4e2dd3982c816","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2026-10-06 10:48:39.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"f6c430b75874e9a11e5d094393efca226dfe1d01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 11:05:00.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa5d796244d48fd9b6cfa1740b8393c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/153730a4ebd64e20b8b52c7b05916d3b : SUCCESS in 15m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98bebfc891d9467dbdddd098889a5728 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"118a6d8d49eac87bb814d9b0f6e578149250afdf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-06 11:27:59.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1eb8e9d9a08842bf99160795199b18ee/","accounts_in_message":[],"_revision_number":13},{"id":"f7a7ed2da148174f0676f423fbb80dc409d63ea2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-06 11:28:45.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5847/.","accounts_in_message":[],"_revision_number":12},{"id":"754c77d68997d432eee1019c1ddbab105c77f506","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-10-06 11:40:25.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa .","accounts_in_message":[],"_revision_number":13},{"id":"1fb9ea1381903775a055db0226840a64382fd954","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-06 12:08:34.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1eb8e9d9a08842bf99160795199b18ee\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0ae072d955ba4dc7b949931c2a5dba12 : SUCCESS in 40m 44s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74703b7ef09f49f39e5e47474396f12f : SUCCESS in 34m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab662764618f4767825dcd4dcb9abd84 : SUCCESS in 45m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5c65a6d4f54427592cfd07631342953 : SUCCESS in 26m 43s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83c76cfe4cad453788625453c90932b6 : FAILURE in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"02e0246b88557caa0421a6f7013f1feabdc22556","tag":"autogenerated:zuul:check","author":{"_account_id":29716,"name":"LINBIT LINSTOR CI","email":"linbit_ci@linbit.com","username":"LINBIT_LINSTOR_CI"},"date":"2026-10-06 12:17:00.000000000","message":"Patch Set 13:\n\nBuild failed (LINBIT LINSTOR CI). Comment \"recheck\" or \"run-linstor\"\nto run again.\n\nhttps://zuul.linbit.com/t/linbit/buildset/5032adbb2505436eb6f6b2e10f0fb668\n\n- linstor-cinder-drbd https://zuul.linbit.com/t/linbit/build/ee116f1d56954ac6974b30936fe2ba03 : FAILURE in 1h 35m 06s\n\nLINBIT LINSTOR CI tests the LINSTOR Cinder driver and does not vote.\nWiki: https://wiki.openstack.org/wiki/ThirdPartySystems/LINBIT_LINSTOR_CI\nContact: linbit_ci@linbit.com","accounts_in_message":[],"_revision_number":13},{"id":"2074ff9b42eddab02af45aee37e5993a4af3c48b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-10-06 12:37:52.000000000","message":"Patch Set 13:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 49m 5s\n- FC       : SUCCESS in 0h 50m 55s\n- NVMe/TCP : SUCCESS in 0h 50m 57s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/973593","accounts_in_message":[],"_revision_number":13},{"id":"36b7bff9346659e6f063a2ef5b5370c10263077c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 13:21:40.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3588b8a6794a66a1f2c407f88c6b7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033e182107c545aaa3fddb4aca8f216e : SUCCESS in 8m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba8a67de7b954c9db7c3c9ae4c42d77c : SUCCESS in 12m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbd0eed37f7f44138db9f3d123207898 : SUCCESS in 16m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1baa59a363a4424188b9c1f6bfdb8011 : SUCCESS in 15m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b27c32c2a495440bafce175a2b91cc5c : SUCCESS in 1h 14m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ec4a13e65cd4bfb880847a290249497 : SUCCESS in 1h 11m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/254fa685190148e2a23723a286c5a986 : SUCCESS in 1h 32m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d218b622b0044bea1383dc0a5e66e10 : SUCCESS in 50m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0cc9da377344c71b1fc87861e378470 : SUCCESS in 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/055b48708c184519bfe3fca7f8a95d85 : SUCCESS in 25m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3b8e1c2bee2743f08e7e4c9485789d3a : SUCCESS in 6m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a028881c6f014a5bab3a1b712c59af98 : SUCCESS in 7m 01s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7a4e9de2de85466eb3a7f0c5f49bc759 : SUCCESS in 10m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/88a55db81c874ebdbe5c47099418b114 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7fbe41262416472488404f0d63312fe1 : SUCCESS in 7m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c647692df9974092afd6986ba6d0a673 : SUCCESS in 8m 11s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6acce968913b4fb4adbebe562d1c4758 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3d610c2cb57d4eb5ac89344f4a601b78 : SUCCESS in 9m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8bcae7aabada4d948eede6c817350d49 : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e454bf71e63d437ea60806e0eee1a9a3 : FAILURE in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aee7898a41d6484c8102adf3957f4313 : SUCCESS in 1h 58m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8fb4664e01ca4c28a723a3e570654844 : FAILURE in 1h 42m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f611d44f1924124af9072e69c15a555 : SUCCESS in 1h 52m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e80a945ab88e436bb6345dff3c9dfc09 : FAILURE in 2h 35m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3410090f3850452aa6d034678a59bd1a : SUCCESS in 1h 52m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/998c317908ff449e8a866cd627cd4ded : SUCCESS in 34m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4d2dae784b84244a7ab0f2480bfeb99 : SUCCESS in 1h 15m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5d91f4e42524b48bc9eae011bd4ca4e : SUCCESS in 2h 02m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e4477ec578f04319a25907ad2dc90298 : SUCCESS in 41m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b8eb8a29a254926a58d990168174ec8 : SUCCESS in 1h 49m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5479b39720bc450b9c3bc17f5cf5a20f : SUCCESS in 54m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"5629569eed0551048faba23e0b44699cd139f9eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-06 13:31:31.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/3880aeb4b3244084963fa10f4c87cb2e : SUCCESS in 44m 52s\n- cinder-storpool-tempest-iscsi-multipath build/2e08b7ac166b4a1f8a14b8cea7090231 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":13},{"id":"47e58107f994e9ff4e1049be099010191c491548","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-10-06 13:36:25.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5682/.","accounts_in_message":[],"_revision_number":13},{"id":"0d14e765b97fe4300edfd781fa25eac1c05a6d83","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-06 13:55:46.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_10_2026_13_49_52\nLeave a comment with","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"4471ed88f76bbc824ff236545c3adc0ce121ce56","revisions":{"1809b678203c608f6522ac96d725450732313bdc":{"kind":"REWORK","_number":1,"created":"2026-01-16 10:04:44.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/1"}}},"commit":{"parents":[{"commit":"58e4987b543e6b6a954a7c8537b5f5ae33f58a3c","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58e4987b543e6b6a954a7c8537b5f5ae33f58a3c"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-01-15 14:42:54.000000000","tz":120},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1809b678203c608f6522ac96d725450732313bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1809b678203c608f6522ac96d725450732313bdc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"58e4987b543e6b6a954a7c8537b5f5ae33f58a3c","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"},"c26be820268da61c27e7ea42ce2255653787822c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-02 07:57:13.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/2"}}},"commit":{"parents":[{"commit":"03e4b5c299a4381ff53ee9b7d3b0fb88275ef093","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03e4b5c299a4381ff53ee9b7d3b0fb88275ef093"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-04-01 14:53:12.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c26be820268da61c27e7ea42ce2255653787822c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c26be820268da61c27e7ea42ce2255653787822c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"03e4b5c299a4381ff53ee9b7d3b0fb88275ef093","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":2,"change_status":"NEW"}],"branch":"refs/heads/master"},"54fda3aba03ec741ef76213ef3dd9be6d34ffd10":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-30 14:22:15.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/3"}}},"commit":{"parents":[{"commit":"9351c9aed853191fcd2edb99aa1523c897d30fcc","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9351c9aed853191fcd2edb99aa1523c897d30fcc"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-09-30 14:20:09.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54fda3aba03ec741ef76213ef3dd9be6d34ffd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54fda3aba03ec741ef76213ef3dd9be6d34ffd10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9351c9aed853191fcd2edb99aa1523c897d30fcc","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":3,"change_status":"NEW"}],"branch":"refs/heads/master"},"5851b7f9ec4c5566d988ba72468370affba76fa4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-01 08:54:08.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/4"}}},"commit":{"parents":[{"commit":"2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 08:08:39.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5851b7f9ec4c5566d988ba72468370affba76fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5851b7f9ec4c5566d988ba72468370affba76fa4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2c6cd0e17bbb60f9c699eaa2cc9e9c7821828baa","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":4,"change_status":"NEW"}],"branch":"refs/heads/master"},"56c295f039c7b7d32ebd49c129a8e901711e26b2":{"kind":"NO_CHANGE","_number":5,"created":"2026-10-01 08:59:33.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/5"}}},"commit":{"parents":[{"commit":"e6acd8df4ebd32b018237583e0c8ced68fb8a82e","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6acd8df4ebd32b018237583e0c8ced68fb8a82e"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 08:59:23.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56c295f039c7b7d32ebd49c129a8e901711e26b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56c295f039c7b7d32ebd49c129a8e901711e26b2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e6acd8df4ebd32b018237583e0c8ced68fb8a82e","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":5,"change_status":"NEW"}],"branch":"refs/heads/master"},"8f1a82b434d322f1c59ad62a192ab00b83343ab0":{"kind":"NO_CHANGE","_number":6,"created":"2026-10-01 09:00:25.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/6"}}},"commit":{"parents":[{"commit":"1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-01 09:00:20.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f1a82b434d322f1c59ad62a192ab00b83343ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f1a82b434d322f1c59ad62a192ab00b83343ab0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1f0e866f1aa85fd35c9fa3ba5f6e0cecdc14fd10","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":6,"change_status":"NEW"}],"branch":"refs/heads/master"},"d7e646bc5fa0118ad601f935ba46a2f70bcdc9e1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-10-02 07:50:07.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/7"}}},"commit":{"parents":[{"commit":"9d120098274a470ef25bf7a67b06dcc179d6e49f","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d120098274a470ef25bf7a67b06dcc179d6e49f"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 07:36:17.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7e646bc5fa0118ad601f935ba46a2f70bcdc9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7e646bc5fa0118ad601f935ba46a2f70bcdc9e1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9d120098274a470ef25bf7a67b06dcc179d6e49f","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":7,"change_status":"NEW"}],"branch":"refs/heads/master"},"01b7a53b1143bb4e72dfccaebfd31d8ec1dbe6d7":{"kind":"REWORK","_number":8,"created":"2026-10-02 08:28:47.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/8"}}},"commit":{"parents":[{"commit":"2b576744798b6e9df42ec177c1d4967ff7d43c34","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b576744798b6e9df42ec177c1d4967ff7d43c34"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 08:28:41.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01b7a53b1143bb4e72dfccaebfd31d8ec1dbe6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/01b7a53b1143bb4e72dfccaebfd31d8ec1dbe6d7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2b576744798b6e9df42ec177c1d4967ff7d43c34","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":8,"change_status":"NEW"}],"branch":"refs/heads/master"},"41117c1e2411aa51e5f61a339f3488c9cad49ec4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-10-02 12:03:25.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/9"}}},"commit":{"parents":[{"commit":"56cd9e4c52a756bab98912da1754ac9a3464dff5","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/56cd9e4c52a756bab98912da1754ac9a3464dff5"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:03:14.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41117c1e2411aa51e5f61a339f3488c9cad49ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/41117c1e2411aa51e5f61a339f3488c9cad49ec4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"56cd9e4c52a756bab98912da1754ac9a3464dff5","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":9,"change_status":"NEW"}],"branch":"refs/heads/master"},"1d3ad63c3369305388f8a05f6def0e61bd5e565b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-10-02 12:36:55.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/10"}}},"commit":{"parents":[{"commit":"8e9192e6d4ec283a05fb890c17276e997ec7a043","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e9192e6d4ec283a05fb890c17276e997ec7a043"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:36:24.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d3ad63c3369305388f8a05f6def0e61bd5e565b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d3ad63c3369305388f8a05f6def0e61bd5e565b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8e9192e6d4ec283a05fb890c17276e997ec7a043","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":10,"change_status":"NEW"}],"branch":"refs/heads/master"},"1683927b6e7afe9edbd72496e6cd2cbd1da98166":{"kind":"NO_CHANGE","_number":11,"created":"2026-10-02 12:41:38.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/11"}}},"commit":{"parents":[{"commit":"0fdc1ea96665097d3c8408a6e3187526d8bd7b47","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fdc1ea96665097d3c8408a6e3187526d8bd7b47"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-02 12:41:32.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1683927b6e7afe9edbd72496e6cd2cbd1da98166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1683927b6e7afe9edbd72496e6cd2cbd1da98166"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0fdc1ea96665097d3c8408a6e3187526d8bd7b47","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":11,"change_status":"NEW"}],"branch":"refs/heads/master"},"284cd93d9b367e7ac134372d92f0aa8c946072ea":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-10-05 07:04:56.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/12"}}},"commit":{"parents":[{"commit":"5f0fbce1942514a98a330a662dfb9a017445119d","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f0fbce1942514a98a330a662dfb9a017445119d"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-05 06:54:30.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/284cd93d9b367e7ac134372d92f0aa8c946072ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/284cd93d9b367e7ac134372d92f0aa8c946072ea"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5f0fbce1942514a98a330a662dfb9a017445119d","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":12,"change_status":"NEW"}],"branch":"refs/heads/master"},"4471ed88f76bbc824ff236545c3adc0ce121ce56":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-10-06 10:37:37.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/93/973593/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/973593/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/973593/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/973593/13"}}},"commit":{"parents":[{"commit":"acbe972d5a8e2e648dce55cf3a5376767e97a6dd","subject":"StorPool: Do not use the option \u0027storpool_replication\u0027 when creating or retyping volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/acbe972d5a8e2e648dce55cf3a5376767e97a6dd"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-19 20:52:16.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2026-10-06 10:10:44.000000000","tz":180},"subject":"StorPool: Use only baseOn when cloning a volume","message":"StorPool: Use only baseOn when cloning a volume\n\nStop creating a transient snapshot when cloning a volume to a different\nplacement group.\n\nChange-Id: I6e53df78bcd9226898fdb40d2ae420cbe1f6c9aa\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4471ed88f76bbc824ff236545c3adc0ce121ce56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4471ed88f76bbc824ff236545c3adc0ce121ce56"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"acbe972d5a8e2e648dce55cf3a5376767e97a6dd","is_merged_in_target_branch":false,"change_id":"I45d32892e7712cb3bb4e52cf4547691505b01d01","change_number":973592,"patch_set_number":13,"change_status":"NEW"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
