)]}'
{"id":"openstack%2Fcinder~974292","triplet_id":"openstack%2Fcinder~master~I4904571a460ad12b993214bc82aabdf087a600bf","project":"openstack/cinder","branch":"master","topic":"pbha_merge1","attention_set":{"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-06-08 14:56:49.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-08 10:00:10.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-04-29 00:19:56.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-06-08 10:00:22.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":["ibm-svf-cinder"],"change_id":"I4904571a460ad12b993214bc82aabdf087a600bf","subject":"[IBM SVf driver] Support PBHA clone-from-src","status":"NEW","created":"2026-01-22 06:04:03.000000000","updated":"2026-07-13 07:43:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26b401c37133d0ed5c1c1f55736b75bf78eefc97","_number":974292,"virtual_id_number":974292,"owner":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-08 13:33:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 06:07:41.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-01-22 06:33:51.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2026-01-22 06:55:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-22 07:04:56.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-22 10:08:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 10:15:45.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-01-22 10:49:10.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-01-22 12:54:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-01-22 14:11:47.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-01-22 15:32:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-23 09:42:14.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-04-10 07:00:41.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-10 07:52:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-10 11:09:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-04-14 17:23:27.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-28 23:41:00.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-04-29 00:19:56.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-04-29 00:34:11.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-06-08 10:00:22.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-06-08 11:49:10.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-19 19:55:32.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"5d723cc5595c616187bcd3961e2cb814b7aa1160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-22 06:04:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d076153b10fc759258c1711a8c14b208e00c5a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 06:07:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d4e67910a75e4225bcc1f5a26a3b4c89 : NODE_FAILURE Node request 200-0000623314 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1b36d644fc1d2d1c9844724284a89fe5591f46a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 06:55:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9efe2dd2ae4d4f048d6eccfcbc8c6966\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/36655f63bddd461e8c7e1a44ebea1b8a : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/264040cf87c74f1a9b37805f09034430 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb0e5b9cf18b41aabfdd91872d01429e : RETRY_LIMIT in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a6cc85f8c54b47a18583f8cd254ddc25f70fb18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-01-22 06:59:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d797e62dd8390aa52e621e717e651b14fd76115","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-01-22 07:01:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/46b811e8a20e480abd4605ffb409432e : NODE_FAILURE Node request 200-0000623318 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9259d3a4dcc53e2456fb0dee785f428406133fa0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 07:04:56.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":1},{"id":"672ddad32672d199d935938487305152adafa4bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 07:26:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0345c3fbc7aa4d94a368547e58054468\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0899b567377a4238b30e07d2ed209bba : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3a3ac291477347329cf67a4e51855501 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1feecd76f54345798867debb9d35c269 : RETRY_LIMIT in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3f7f329050768bedf85d1cf0936a52c6d331023","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 08:01:21.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":2},{"id":"c9749f6c5e1351d2e2e5461babe292af577d062b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 10:08:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1ae1a92e7344fb821ead5f282e905f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf65370be204a659bf99e4edb235f7d : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52a1e4cc26f24853a068ad7970907ac6 : SUCCESS in 11m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/759980dd7c1043b9829a32d49c7c5a2b : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5066289963f747788214aee70c004670 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/74805c69472d4afebd74f13a44e87323 : SUCCESS in 34m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f58a7389d87432a9d0455e29b4dfbe0 : SUCCESS in 42m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f03f01278f444426901212c161e883ff : SUCCESS in 1h 31m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/543b278ea0cb4bbd972da8ec2a6bd7e1 : SUCCESS in 35m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d894122274f240958b751f1b3d82a728 : SUCCESS in 16m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f1793ee1d363487698c89bb8fbc035c2 : SUCCESS in 19m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b85c28cf9eb433793acf4b3eca7612c : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7896cfb5e7f4d66b6f41fb1bc06c9e3 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a11e936318f4a28960ae89f99427fc3 : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6a3f3a7aa2364b0085149f15c80663b9 : SUCCESS in 4m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/775797a847b64db1865df1c8a2698272 : FAILURE in 39m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e92536f5c06a438fb4ffd02f5034309c : SUCCESS in 4m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/97ff539cd15b47dabd5c6da8373a80f1 : SUCCESS in 1h 47m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2660b3bf45eb4513ac49dcd9b2c889f4 : FAILURE in 3h 03m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c706b7534bd54531bf451f63f715bf49 : SUCCESS in 1h 50m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0bc92a14a3f7457caccdd0f9b366fbe3 : SUCCESS in 1h 50m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f77f86c7855049199e0d1e0d0149faed : SUCCESS in 19m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2b4780a384f4658980db074996a4a2b : SUCCESS in 1h 15m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4bcf5f4e91b54122af3604da692b4e07 : SUCCESS in 1h 41m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83b909979f5a40bf9fb184fc97bd0839 : SUCCESS in 41m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0943be70e347495d93b216e39aed724b : SUCCESS in 1h 56m 28s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/c7cedafc4810426bb2c21f47871b9223 : FAILURE in 2h 22m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ca25b4286ee4d98a944ad32927e3197 : SUCCESS in 1h 31m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7954d5adb7554459a58de3ad1a3ba99e : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ac87c104235acd7fc95ae2e961b65b6d44c95805","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-22 10:15:45.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_01_2026_10_09_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"48e398f571db9657a5a2dd11f2cbf0b9a63aafc5","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-01-22 10:49:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/970deb8fe8fc427cb369314600bb2341 : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":2},{"id":"181545bd63a7dad9e7b4a23bb7aaa08cbb988ec9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-01-22 12:54:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/974292 : FAILURE in 35m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5b5984dcb6203c755341a60ec1c9c089c543bcaa","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-22 14:11:47.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_326/index.html : SUCCESS in 105m","accounts_in_message":[],"_revision_number":1},{"id":"bb4f66bccb03e87495b0d359b24f2da1d8cf340d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-22 15:32:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3da419971dde4fe0866250f4f7998197 : SUCCESS in 46m 21s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/aaddb203c86c404e84344e8fdd7bd1ad : RETRY_LIMIT in 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6687a9226f75f46a749d5814d31ac017962d3245","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-01-22 22:25:44.000000000","message":"Patch Set 2:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_331/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":2},{"id":"707c1fafd2db68434c575e7d19bdcf52e8904542","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-01-23 09:42:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-974292-2-fc : SUCCESS in 2h 19m 42s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-92-974292-2-iscsi : SUCCESS in 1h 44m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cc5265c15d817c9f2d6dd5106b195514d93c7bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 09:44:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a28fa66395e4fcf1105a98399532bea3118035cd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 09:58:19.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"add609fb905640fc09fdbdda6dcdc7b260c7c058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 11:57:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4ae73239fc41379bb8794929ea0c84\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/857873d0d5cf4784bc89b6e61166c537 : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5e03a64ea8f48efa7fd18378c105ac2 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ce6689b9966412f93dd0069480682ac : RETRY_LIMIT in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d199f5c24922ef6e737c05f2459e428958a5b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:19:26.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f49ba796d1a49b8a6d79b79fc088b1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3bb77cd8e6439899c3e8f92274bfbf : FAILURE in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abd8acb704c041dfaf34beb09dbd8758 : SUCCESS in 10m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43c35289c1394f00a7467f16043b6f53 : FAILURE in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9052b64ce834fc9abc3bea5211013a8 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/34224c8d2305455ea6e758332928f19f : SUCCESS in 36m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c68968bcc0994cb983a2b20a88c3c9c0 : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d1421828e16490f8688c3602aab8d8d : SUCCESS in 1h 48m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfd40e574cc9492bb30f0ddc489371c0 : SUCCESS in 34m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b78ad0f0c05a4a559c6a7b930bae30b0 : SUCCESS in 17m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16e99ad0fa8747f384125fce0051a9aa : FAILURE in 26m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/12b68221424e4d4295398b02fa746ee8 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7572eca69eae4655b0f8e8f3aab23bff : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aaf119dca5f94e809747c5c44af27df0 : SUCCESS in 7m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/beac5e53e2044ebc9657bff100db27f6 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c37f17b43ac4213843f6fdabedac417 : FAILURE in 43m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b70680e4f33845c5afeff0343d3c9662 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d28cd6611d404791b2773d9e8c14d971 : SUCCESS in 1h 48m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/76361c45e32844fd90d41a9591a89c6f : FAILURE in 1h 56m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/83e632f4e5a5458dae5e9b99ddeb6de6 : SUCCESS in 1h 47m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/77ab3ab8fd7743438637c8d44d882a92 : SUCCESS in 2h 08m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/49221a52b3b04c28831876c4f0c6bf9d : SUCCESS in 36m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/92fa8bc4755946cca9882bbc9dce9b44 : SUCCESS in 38m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7791efb89bf5429693faf6229f0c4100 : SUCCESS in 1h 42m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/26adb0962535458abd34ea3df732cac5 : SUCCESS in 40m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79b50dffe3c4473f9b7f52650fe1d6f0 : SUCCESS in 1h 49m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b0764be49c6c41829ba561ef09651fd4 : SUCCESS in 1h 16m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/806a92467c9842699da0039ebb49ce74 : SUCCESS in 1h 47m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96ec8c96bac74a94aed864a39b2e85db : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e855c2ba92825fc20ecc790a9e0bbbff673b74e3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 15:02:01.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":3},{"id":"4fafe2dcb7514e159e3671b2931f756a8f8c6a79","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 19:01:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2bf6c900e1db457bb25170ea278baf7e : FAILURE in 46m 49s","accounts_in_message":[],"_revision_number":3},{"id":"55ee7bb39fbe16cc03c4b52bcfb72d890419c092","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 20:31:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/974292 : FAILURE in 39m 18s","accounts_in_message":[],"_revision_number":3},{"id":"fee20c7f22a7a67156fc80640e3c4b7aa40fe74f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-06 15:07:23.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 938771,101","accounts_in_message":[],"_revision_number":3},{"id":"97acd5d9f19fdfd995c6a12b3caa76672b368ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-04-10 06:17:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"01ca10d0445d4a1dd8fa8061ea7c1cdbce5c3c01","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-10 06:25:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/800e493c9d7c46dda763eb92fba47fe4 : RETRY_LIMIT in 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c3eca89370a64891881743a5e62a07cf : RETRY_LIMIT in 1m 09s","accounts_in_message":[],"_revision_number":4},{"id":"3b94ef1513d842a67dbb057da51c4b766ea00a9f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-10 06:28:56.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":4},{"id":"9ab020312a8cd26dfb1aa5f7850b19086ca2fc54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 06:47:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52466b077954717bbd0f67c363684a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6e137f99924a4b7d9098b73541faf81b : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/04e657cc7f164575b5072d639aa5959b : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/37777367198e4a50a21c0351b7a78e1f : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"976b18c0c9e63aff90f7b8ab11452c95b5442b4d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-10 06:56:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5d133299cb7a4a448284676cdf6c7059 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":4},{"id":"dbb246574504eaee367b985d1dbfbcbdf545bdee","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-10 07:00:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/6d0fb78f01d24e079aaf2dc74cf95cc4 : FAILURE in 29m 03s","accounts_in_message":[],"_revision_number":4},{"id":"68454ef9221373be30fa3f4b678415d3f58c6caf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-10 07:14:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/974292 : FAILURE in 36m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e6c37b0e4e2c5059c3d3477f86d398d043f299e0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-10 07:52:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/8f334fa9fd3e4eb1aeb095060b1fec50 : FAILURE in 40m 10s","accounts_in_message":[],"_revision_number":4},{"id":"585a74395c1311be4a6d896faa2b576df7ed99d4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-10 11:09:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3751/.","accounts_in_message":[],"_revision_number":4},{"id":"233bdf53ce5f90e2c50ad51451abd173ed3f804d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-10 11:25:04.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3575/.","accounts_in_message":[],"_revision_number":4},{"id":"5188be3ff468acb6e3fa597da73d8846532137b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 12:43:46.000000000","message":"Patch Set 4: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48280199a32d4d4f96e31be646993299\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a9b38e621f49598df37f2a5c1a84a8 : FAILURE in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/652d883637224a45ad745b6231148ea0 : SUCCESS in 10m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af2cd8e2a5354cbcb64528357dcbcbac : SUCCESS in 11m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbe8aca4fc2d48daa1d0a933e847ef79 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d276368b35e4beebbf23bfacdeca9ef : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/c6f7d2e98076464887c704fa2b788bde : SUCCESS in 31m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77024eb887a5498ca6a5b0665082c131 : SUCCESS in 1h 05m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef13f399d5c941878ef88d75581a04b4 : SUCCESS in 1h 37m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9eb5ea2cb9e469392417853bf8c6b62 : SUCCESS in 55m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ac20a884a54f939b813305367806d5 : SUCCESS in 14m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/616ded6a738e4c8a9515846e4d4c6ae8 : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c07d4b495c1409384df7cb02b0aacdf : SUCCESS in 2m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/014cfa6dca6749dcb3c6436c45e3f980 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f5461fc875c446e8bb424b463794929 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4f2296fc536f417194d6c9a1a7982614 : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5d8fe581f224f918bd1b3446fea7ba5 : FAILURE in 19m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ec678eb9742d46ed8439421ef5118cdb : FAILURE in 3m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed6b41170557407489309b82ba540b4e : SUCCESS in 1h 02m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c759d9971b074c179523a86844665d01 : FAILURE in 2h 35m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/da4bec93541948d8971d2a4182071356 : SUCCESS in 1h 53m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/be4c88ef9fe94339ab4424dcf8fc0ee9 : SUCCESS in 1h 53m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9993aa1988804d0f8c733692e9479840 : SUCCESS in 33m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/81d426f35cb345b38d6bab482ef770c7 : SUCCESS in 41m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92322256e32e4b61b93f9fa0ffe18675 : SUCCESS in 1h 34m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f4b11cd7732a4714a2694d867fd99597 : SUCCESS in 41m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6ebcc89a8b1e4540ac0fc998af4b22c5 : SUCCESS in 2h 00m 26s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bf5d6d1684544d628c4cd0a2e6e9951b : SUCCESS in 1h 35m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07edd9bbe6994540a7cc1615a4a75df5 : SUCCESS in 1h 24m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22758be5bbe0408bb42eb408be9da373 : SUCCESS in 1h 00m 59s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":4},{"id":"28f3e88519d49f65bc030e7ecf4467a5ebfc34ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-13 07:27:00.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f1dc702e45a359789ae5a8f9095fc42c39b4515b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:43:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8f7a2ba4b3164792a26f9c56be1df7dc : RETRY_LIMIT in 1m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8c772854650f48618f9add7480d895c2 : RETRY_LIMIT in 1m 02s","accounts_in_message":[],"_revision_number":5},{"id":"6b0fcc13007d355ba2f1b7aec746ec4a378fd0cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-13 08:50:19.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":5},{"id":"e2855b3410be51e9028789e740be7d9069432ec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:03:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2dd578ccd1348a088cb6e3e81d252af\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b17e971a009e4699a39b085da41703d6 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7774c5cc462496abf0f0d523491d6a7 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/00613f6677054dba8e0d5ebd3238b397 : SUCCESS in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb7eaab0f6e65789fe43c4546e8ba6c799006449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:11:50.000000000","message":"Patch Set 5: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73efecddae8243df9aa8f3c3c31884ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1eacadec2924481bc595be6662573c8 : FAILURE in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0686aa115fa748b28489cf9163a9f657 : FAILURE in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/393b1763f4ee4f62997a3df09a8fb007 : FAILURE in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/745537de0d574fad832ae02cfd18c261 : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fa156af10a40e1b92baa3b2c58dd68 : FAILURE in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee4bec285a14cdaaf03e675bed53a02 : SUCCESS in 34m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc09bb284c674e5da989131b9d6eb605 : SUCCESS in 32m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56b4dfdf9c4e448cbee7ab493b1e874f : SUCCESS in 1h 43m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cce41aa1b2e494cbfa350a2ecd96096 : SUCCESS in 1h 08m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a46dd8a26f74e5c94334f973391aafe : SUCCESS in 22m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54843626f6d94e059cd60c36d47b080c : SUCCESS in 22m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/634939c3ccfb4754ad20c5b57842a0a2 : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7969af398250411a9e6a82676247f7f2 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5256ad2f898b4bacad0f130080376b6f : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5c79b69de1f64a8490fd19abf4fdf791 : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/43935622f2e1435da35486accf917fd4 : FAILURE in 25m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37066f1c0c0d4ffc9e5ea6b1f7d222f4 : FAILURE in 4m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9fca9d4c2f0b4236b5ccd9ab4b94d7e5 : SUCCESS in 1h 48m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1c16f70c3c994d5abdd3163095f006de : FAILURE in 2h 42m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b6b7c1624bc44f080150f5629ded9fe : SUCCESS in 1h 43m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fada9363f535436e97e35f7bdbb2f1f7 : SUCCESS in 2h 20m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1fd3a482f4d04f04a53783e542e71c12 : SUCCESS in 36m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b32e9614933644c2bffaf83840de4d53 : SUCCESS in 1h 21m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5cfb3a53f89f4f73814e9634e08670d3 : FAILURE in 1h 40m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/379dc8aa14d849ae863f87a8aababce0 : SUCCESS in 47m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc1dae0630b648ae9f2eb1cddb8cd682 : SUCCESS in 2h 08m 12s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0057a1130ac3464db9ffe58483d6669c : FAILURE in 2h 20m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/609a3fe7d8f04439b70bfedf6dc9ad61 : SUCCESS in 2h 00m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/293bb4ed3f394a0eb2548f9239aeac37 : SUCCESS in 59m 37s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":5},{"id":"0da63991cd8da2135eb42175734abf9b301fd3c6","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-13 11:42:58.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/967f69d10fd946939826200b2acea241 : FAILURE in 27m 35s","accounts_in_message":[],"_revision_number":5},{"id":"b3512f966955aea9f58190bbd686b5311c85589d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-13 11:45:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e9cbe862a448478f87da523aa789698f : FAILURE in 23m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b43aaa27d3f2d702c4b450c18cf14eb8dfddbab0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-13 15:59:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/974292 : FAILURE in 37m 49s","accounts_in_message":[],"_revision_number":5},{"id":"18fd7fc1a4f44af15994f010ad4906ee752342bd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-13 16:51:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/39ad11d9c9f84b93833acf58f4c06cba : FAILURE in 36m 35s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"19decda1b53ca61935fbe2bce06f4d9bc6ce732a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-13 22:59:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3598/.","accounts_in_message":[],"_revision_number":5},{"id":"aa7d508adf59b74ba066f8374cc2518e18318893","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 00:34:51.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3774/.","accounts_in_message":[],"_revision_number":5},{"id":"54efbe9d22d122b657c596479f4125640b3e3fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-14 11:04:51.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e2d72af06ce38035c884d2549c034b4b7c8d799a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-14 12:12:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ec3969dfdde24e18b945b053661596e2 : RETRY_LIMIT in 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f8b5c9d7ce1f465899657298c2f06ae4 : RETRY_LIMIT in 49s","accounts_in_message":[],"_revision_number":6},{"id":"ffe5806bc7ba6211dc5afdeccfe995d825055c99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-14 12:25:41.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":6},{"id":"dcaf697f0dad06a197094c074950c4fd4efec0ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:45:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0973ff54a0824ce3b0b58ee4ac3ee344\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3e5f41b99344ce4b29aa8ae51dc1785 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ffb2b7015a24823861f9fb7e8fe77d6 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0539a09f7ed4cda9e19824ecac5bc92 : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a6626def12584bc59ec7314f4b444a9540cb9b36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 14:56:43.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3624/.","accounts_in_message":[],"_revision_number":6},{"id":"0965e235f47b73f191350d1f228ab70d5585eb3e","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-14 15:57:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/78fcae88744f4a4ea0ad5ab978ba4e3c : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":6},{"id":"1e5101d81932cc8c6c5869e7470baba16b4d8ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:43:23.000000000","message":"Patch Set 6: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e75c7a5bb0940ddb395e065d4df97df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70eea02652894976aadda5ba1d57a19f : FAILURE in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8a590cadf2c4d4d847a0b8eb4f07651 : SUCCESS in 16m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dde4ab074707425b9b4ec4b805c59cad : SUCCESS in 11m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/86522aeb4b204255beffd6682d2907ab : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/549a83a1a7b44496b9e579a4a7540fb0 : SUCCESS in 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/cbbec9be6e78424b86d608f1b053bade : SUCCESS in 33m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66318fcb55d64380a1776f3a90925a23 : SUCCESS in 37m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e335ead80bb4f29b8f968acbf866fbe : SUCCESS in 1h 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3be405f0f374b8a853f853e1c5561f6 : SUCCESS in 39m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20da43f33bb7407e9a28318dc3ab6441 : SUCCESS in 22m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bcd2d69534fa44469bebf6b9eb566650 : SUCCESS in 21m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8fa7f953944646679b06ec18bafbb35a : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/80b67da50cac4d5484b67011ce7cb873 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/49d08ab008a24014967bbcc82af655ec : SUCCESS in 7m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e4cdb247227429abbdb1e7ad7f83f91 : SUCCESS in 7m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fe9aef4410a949d0bee8fa60fcd280e3 : FAILURE in 41m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d78c8a748a4a4a74aea6c9b752a8f230 : FAILURE in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e2caeeabe134dd5a45a260a4eb6f252 : SUCCESS in 1h 42m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cd150bcfe0144d92bd6b5830bce2a890 : FAILURE in 1h 38m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e788fb9f665448a688e4da9ec8ec115c : SUCCESS in 1h 48m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5146cdd01fc848b085cfb5bf76e2a596 : SUCCESS in 2h 00m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a4fafa5ae57d41c9a2f703a150cccc3d : SUCCESS in 31m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e37e3096f0d4dc0a37d7d9e330b7951 : SUCCESS in 45m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22ba4a2599754402b5bbf1bf9dbc44a0 : SUCCESS in 1h 35m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ba1373545e84510959ef2a32ba408c3 : SUCCESS in 28m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/956db6dac3ad4682a593af600212cd05 : SUCCESS in 1h 49m 53s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8d5dc41348a442f79895982892be1fa1 : SUCCESS in 2h 15m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9564cacd7df04022a4dedc9e45b2573b : SUCCESS in 1h 37m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/377d93a41e17471881e425428fca0fd1 : SUCCESS in 1h 02m 05s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":6},{"id":"ea1f47305cb1aa6d53d2f6bd2cdddad4e27f5d94","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-14 17:02:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver build/d45ee4951d894644b07f8ed912b67961 : FAILURE in 31m 14s","accounts_in_message":[],"_revision_number":6},{"id":"e3a3ab90f7b707fd5d45dabd6715417d917a42eb","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-14 17:23:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4adccb1bb426d09123e4f155a2d7e15e35f00856","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-14 17:59:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/974292 : FAILURE in 35m 07s","accounts_in_message":[],"_revision_number":6},{"id":"68e7fc8a901e65369ffc64574f826ffd3fefb03d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-14 19:42:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c97503f3126c4f348ea982bf38dd3461 : FAILURE in 36m 49s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"d6d752b83a64cfc3aee05ccbc650d9f9274843ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 12:32:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3800/.","accounts_in_message":[],"_revision_number":6},{"id":"912dec72794093e754fb20fedd04640262272ba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-04-28 17:53:21.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"8bcf2ba7be81f563c701c550a22545ee1fb0ce1d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-28 19:29:57.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":7},{"id":"a0ed34adca184c29d013d30c2a53eb99930bcd4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:57:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e718c04c258e426cb53ac0d943d08dbf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e026a21191c64e788eb32bca4f2c626c : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/231b934eb87c4eabab93100b08e42e98 : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3e3ec2c8d02f4da7828c46ffbd5bc23c : FAILURE in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"03fd3f981e2c708adad1a1099a78a07044c1c35c","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-28 22:45:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ec53e3d098e846d3a8af539134b25a54 : FAILURE in 23m 35s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"2aa5d918bd7d53b4841283f913ba3083ba1fc125","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-28 23:41:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/a4d1ca7f7a964e81ad3e586c081a8f49 : FAILURE in 26m 52s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"a13c543fb2f2e5eb86c383c9e1d7988945c5250f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 23:49:21.000000000","message":"Patch Set 7: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36147f0926b044f9bd23568810cf3d92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28aa8a4142934c42997cda5ed55ae54e : FAILURE in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7cd8139e8b44c67a14b963f13bd32d7 : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/623bd9a2f8f248e5b6da36b4ebd54ed1 : SUCCESS in 11m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a2ed51714b940559183892012039415 : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053cc4c5ae494ddc9b22cdd16754a87a : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/9b8d0ff0becf4138bb04b1dc743ac700 : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29e57141ee3d4468ac188a7ef5feaf9c : SUCCESS in 1h 10m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/245ac0b6b8904e13853d1333ea8ba31b : FAILURE in 1h 25m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91ffcf8d269b4231b95f8c589aadd57e : SUCCESS in 1h 00m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6397f704082e43729d7a7d31bc56dfd1 : SUCCESS in 23m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a80cd0ac642b47a0a35ab1e966035312 : TIMED_OUT in 41m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab55b70e5ac54a9fa4899468729f7d2c : SUCCESS in 5m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee2bba0120284a36914f19ef09f6430e : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/437cb1d9cf524b459662360f49d76bf2 : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9b97330b2e5c471f82b7001ffd5b89ca : SUCCESS in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6fe3f24810b14fefade3e98310e3eb30 : SUCCESS in 42m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2984ee134a9a41f3be2eb55ab320a1dd : FAILURE in 7m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee9fe76986294024856aec27e3e3204d : SUCCESS in 1h 44m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/753c0d70e8cd4d319595d5c1a0581f78 : FAILURE in 2h 15m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20cbc0aafc7e4cf49debe86e4af2bb0c : SUCCESS in 2h 34m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0a10709e4b0b4c069b7197f14736d0fb : SUCCESS in 2h 09m 50s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/48466620b3ad4fbdb7ff8c2b27f9bbc8 : SUCCESS in 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/293cbbc02b31406ca97a7ad168f9980b : SUCCESS in 1h 14m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5792664767e4c7b97d817a64bf7f2c6 : SUCCESS in 1h 43m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/01614eb65b234737a808a717f0f209f0 : SUCCESS in 47m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f80ca2ff49b44b5ad47961627a49eb4 : FAILURE in 1h 50m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d95e496f32c24ff3bbebdbedf218cf96 : SUCCESS in 1h 20m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f471acc53c64ed5b654e8114d5b9112 : SUCCESS in 1h 08m 03s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":7},{"id":"2f399baa1977ea28ea636be47e9d79675a9a5cae","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-29 00:19:56.000000000","message":"Patch Set 7:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/974292 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"f1dc228cc45618d3a3c82115d86b256c35caf26b","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-29 00:34:11.000000000","message":"Patch Set 7:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/974292 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"e87c78c68bb4255e07472ff0c1d0647edc818ed2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-29 04:40:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3995/.","accounts_in_message":[],"_revision_number":7},{"id":"56e8443bce96a4bf3a99dd57d71cd823d7b3273c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-29 04:47:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3819/.","accounts_in_message":[],"_revision_number":7},{"id":"f52a4668accbf70fe15dd10506354feeef763714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-04-30 15:08:44.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"6b524dc1a95d6c09c295b77105f2d62a013df44f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:19:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c700bf67d92a4688a5a5ece315a0d710\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b352ba982d3482f80c1984e8c59fffd : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7874b5cacce648c39bfafc9652395914 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/56f43f9a812e4696bfe529a6d1024203 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"01b6e41800c4dcedcb610271593fd8436e84e0d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-30 17:28:05.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":8},{"id":"b716354ad4b7a3fc550f1475e7cbd5e1b65aa6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:45:06.000000000","message":"Patch Set 8: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8ccdcbd9b8d47dfb8c037f944474dd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987c859f4679426ca7f795849895ebbd : FAILURE in 7m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e002277278041fb92a4f9c107d8badf : SUCCESS in 13m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f13756a62e704f3dac8e5f81aa863d55 : SUCCESS in 11m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65143c0a08bc46c286351dc14b442257 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f39f4c352d344f9929d730cbd92904b : SUCCESS in 15m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3f10be2275094c71b26340f22036cbe5 : SUCCESS in 1h 01m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cd901b25fc54dd4bc8493a2632c17bc : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e389615c61504cf6a334756464513663 : SUCCESS in 1h 45m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/250c3da7d33249b09da596165b4fa46b : SUCCESS in 58m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fb32ffc4c40468ab280703227e22c48 : SUCCESS in 16m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/77275c636d7742069d5f42da918cddb5 : SUCCESS in 27m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1d75a8a20c6445c9a51b138cc462f1d7 : SUCCESS in 3m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f1f33f2fc5b49a783fae109ae68a7bb : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6da2afc903a463a96a7e4cb12937f84 : SUCCESS in 8m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f40c78a20c6f4158bb8b31be4ca45eb6 : SUCCESS in 7m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/870926e63521467faf16c7e2721894ea : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e905b4f0e369420eb726f97ac0e4d12d : FAILURE in 8m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29d48d83dc6f44fa9bd8ad108df92238 : SUCCESS in 1h 05m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/263a8fa02556412b871901930222b606 : FAILURE in 1h 34m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25dd3f1fed4d4d8082c8db35c62dd840 : SUCCESS in 2h 15m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4e3ff429aea24c8491a78f6f10cdf8a6 : SUCCESS in 1h 49m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ee3a1c88f3964aafbaa6f9dd887c826b : SUCCESS in 47m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98805144ad604211956dd5f73fbab187 : SUCCESS in 1h 21m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f6f35434a77a41e8b3aa252088ee4f94 : SUCCESS in 1h 53m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d5a0869bfe7498294838aff623ffb5f : SUCCESS in 39m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a5825ee3090e41b294ea3361625ef17c : SUCCESS in 2h 22m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22c6e2ea1b334495ab13711454ef5c81 : SUCCESS in 1h 49m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136064ed490e4ccf967d123fa2b12e37 : SUCCESS in 59m 00s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/ibm/storwize_svc/storwize_svc_fc.py","accounts_in_message":[],"_revision_number":8},{"id":"8564beb37524138028385abd49dbc60e255d86d7","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-30 20:39:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/89820da27ad249c2a7addc38a2aa71b1 : FAILURE in 5m 44s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"2bbfaf95199c89922fc5fd85265459607a6ec87c","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-30 21:24:40.000000000","message":"Patch Set 8:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/974292 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9bc5d894d5ec883662b1368d02aeba1d3e62b846","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-01 03:05:58.000000000","message":"Patch Set 8:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/974292 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"7d4da120279444e3fb0da3c41fa946e015e4cbab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-01 04:25:17.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3857/.","accounts_in_message":[],"_revision_number":8},{"id":"6608183205ba35d146d45d82bd594e6c8174796a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-01 05:53:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4033/.","accounts_in_message":[],"_revision_number":8},{"id":"2fd5d99dfad04de2c3c382e64f0d3bf91876351b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 14:51:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6cfdff25f3c142cc9854d7919c1dded3 : FAILURE in 29m 48s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"ca164ebb7caf01cf6c477e57571c86588c1cbe9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-06-08 10:00:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"bb8a69e08e528905cfaf6079177a95c9fac95248","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 10:00:22.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/861/","accounts_in_message":[],"_revision_number":9},{"id":"b4421c6cf6cd26b7b238101640881accbd518239","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-08 10:00:24.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/974292/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/974292/9/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"ad96b9d5a66c23ac9497d22810bfbd7a3ba1bc1a","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-06-08 10:01:27.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5fa403c4328fb7a1d873c42ce1b843ba8379a898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 10:44:06.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb379ccb462248efb8e863dc12cec0fd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/144091cd675544a384d23b6d49a99d10 : SUCCESS in 15m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8eab6cd5cc494795b14e8de17ef467ee : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b60f2bd5423542e5a9c8e332de697ccc : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d0b2625b228b5b820b124f3506d361e44b2a2d39","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-08 10:50:33.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/c676c93b93f44d519b62b9e041e748bc : FAILURE in 29m 06s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/974292","accounts_in_message":[],"_revision_number":9},{"id":"45f0dd15c20c93b8324c061d59ca827425c79ae2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-08 10:54:19.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/278119faa5864570b3232b0abcff0135 : FAILURE in 23m 35s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/974292","accounts_in_message":[],"_revision_number":9},{"id":"aa7b265c2ad1bf13142b8f3fe16a67ae978c436e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-08 11:00:23.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6f64cc59db1b4bc6ab079c014a2b2769 : FAILURE in 32m 04s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/974292","accounts_in_message":[],"_revision_number":9},{"id":"bda7fd1348d12c1863dd53ca4bbe66a118dd7133","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-08 11:05:43.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4904571a460ad12b993214bc82aabdf087a600bf .","accounts_in_message":[],"_revision_number":9},{"id":"7182a0756be8975856081743fea19a9c13586902","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-08 11:49:10.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/07cc87a8f15541298f7437e2a28b5167\n\n- pure-devstack-cinder-tempest-fa-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9b5a7347281b45a18de82622b1fb91b3 : NODE_FAILURE Node request 200-0000063802 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-fb-nfs-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1274924d8e664c7392817865e5f9cf9c : NODE_FAILURE Node request 200-0000063803 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d5abbd4941214a5aa7d5296a0245bd2a : FAILURE in 1h 22m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/731791a0dbd947568950938a66ff8172 : FAILURE in 1h 22m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/576ef92b143c4392ba00bd78e2c1e6bc : FAILURE in 1h 22m 44s","accounts_in_message":[],"_revision_number":9},{"id":"c922dad99d854bc4023d46113d1a52c8caec565b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:33:54.000000000","message":"Patch Set 9: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34095b684a1a4f899b2218a1593ef41c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaded5746fb8437da62d7eb810cf8f5f : FAILURE in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b7b95ee90974af0b798911875c805ca : SUCCESS in 10m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52c71fc8767d486693567fc6867a0618 : SUCCESS in 11m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5e63442d4044a2b9a9c9537544dbe7e : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a13c1c5fe049e2a044c383b4b46746 : SUCCESS in 15m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/9995feab28284138bbe1958819e55b89 : SUCCESS in 1h 21m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0dfd4c8a6ee44c738a5378287606936c : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f1fd4659bc8431995875fba48401462 : SUCCESS in 1h 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d956867491840c6b11f9855d6ca08d5 : SUCCESS in 1h 07m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e56d072afc483aa3a2910e2a3a1708 : SUCCESS in 9m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e69e314916344bb59b5c49a807817f37 : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/033df9f83df248049c87d2cb2b597812 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e6a190b57a64c0bbed1843b54d1dddc : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20f90281a3f14ab78738f1ec6cc645fe : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/83183ace2a554a4c88fbe13fe2931c00 : SUCCESS in 7m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d970bc554521407fb0c941990f72f501 : SUCCESS in 9m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f743a6cd49d147a4ab8b51de16d133d6 : SUCCESS in 33m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97f881bc26d14591a4becd10ffcb73e4 : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/014864b27c504ecc972901fbf8b85432 : SUCCESS in 1h 46m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b00953571e64031b495bfc691a62531 : TIMED_OUT Host unreachable in 3h 26m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7b6463282a2846bcbaf4602ca112e5e2 : SUCCESS in 2h 05m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b2f0408179954ae6a749889d5ca394b4 : SUCCESS in 1h 59m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d908df714a0f47fa883d2c46359823e8 : SUCCESS in 30m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/543ca6415ff446b7910a9be7870240a9 : SUCCESS in 1h 22m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/506c328c7bbb4d2aaeb0c4071858fcd4 : SUCCESS in 1h 42m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b286873154ab498fb509fc608a07ec46 : SUCCESS in 48m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44610d73b7274d308ec538950aa0c2d4 : SUCCESS in 1h 50m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/706ae2c424054b89bdf2bf80251c4d62 : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":9},{"id":"7d47984f1780b79b2fb5645ff20e5766a1c2aac8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-08 14:56:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4383/.","accounts_in_message":[],"_revision_number":9},{"id":"980ec64fa4ec6772df2c89983fc749a7ba2d1525","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-08 18:06:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/61ac7be6b5cb4916acffd1c0afed3356 : SUCCESS in 1h 04m 06s\n- cinder-storpool-tempest-iscsi-multipath build/a97484adcbf04e5c9abb36b7088d2b60 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":9},{"id":"5e9b1e0170632ff51c5aec766cfb45f8aafac55e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:32.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5},{"id":"51cec6a96a3a1f724e2c0c8919023d71107744d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:58.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6},{"id":"a14c66b73da443ff5ad29ad844413cece2822190","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:58.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"f6b89b027131ee2d3dd9fdead9533fda9b03b0dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:50.000000000","message":"Patch Set 8:\n\nBuild Failed","accounts_in_message":[],"_revision_number":8},{"id":"eb567dd00d6ae1464b7f46b788b82fc7cbd098e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:43.000000000","message":"Patch Set 7:\n\nBuild Failed","accounts_in_message":[],"_revision_number":7},{"id":"608eac3b143f38854c7cf0a58c022fa01f6314a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 11:01:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4207/.","accounts_in_message":[],"_revision_number":9},{"id":"26b401c37133d0ed5c1c1f55736b75bf78eefc97","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":39312,"name":"Nimesh Desai","display_name":"Nimesh","email":"talk2nimeshdesai@gmail.com"},"date":"2026-07-13 07:43:26.000000000","message":"Hashtag added: ibm-svf-cinder","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e0abbd3df52280da3a979420c5ee70120b89c78c","revisions":{"09e90f79e16a275eb71a6c2ada28c819f6b4082e":{"kind":"REWORK","_number":1,"created":"2026-01-22 06:04:03.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/92/974292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/1"}}},"commit":{"parents":[{"commit":"b7f3dab5f0ea22f592638d2666d5d8fd8187f0ed","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7f3dab5f0ea22f592638d2666d5d8fd8187f0ed"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:03:10.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create clone volume from source.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09e90f79e16a275eb71a6c2ada28c819f6b4082e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/09e90f79e16a275eb71a6c2ada28c819f6b4082e"}]},"branch":"refs/heads/master"},"1a936c78f4a6fa05fa4b0638f6f3d74fcc3d1bc9":{"kind":"REWORK","_number":2,"created":"2026-01-22 06:59:42.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/92/974292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/2"}}},"commit":{"parents":[{"commit":"b7f3dab5f0ea22f592638d2666d5d8fd8187f0ed","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7f3dab5f0ea22f592638d2666d5d8fd8187f0ed"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:58:33.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create clone volume from source.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a936c78f4a6fa05fa4b0638f6f3d74fcc3d1bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a936c78f4a6fa05fa4b0638f6f3d74fcc3d1bc9"}]},"branch":"refs/heads/master"},"70b8d25f0040756115b6e1d30a905c04f31f5f22":{"kind":"REWORK","_number":3,"created":"2026-02-05 09:44:17.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/92/974292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/3"}}},"commit":{"parents":[{"commit":"c587d474b08f39a6780bd83dc996083018cb65cb","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c587d474b08f39a6780bd83dc996083018cb65cb"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-02-05 09:42:28.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create clone volume from source.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70b8d25f0040756115b6e1d30a905c04f31f5f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/70b8d25f0040756115b6e1d30a905c04f31f5f22"}]},"branch":"refs/heads/master"},"3488513b2182dccc44e8cd7a3ba62cd275f75867":{"kind":"REWORK","_number":4,"created":"2026-04-10 06:17:25.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/92/974292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/4"}}},"commit":{"parents":[{"commit":"f7f4eb5d79663ceeeb1dd0cf3349af0454f7c2d3","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7f4eb5d79663ceeeb1dd0cf3349af0454f7c2d3"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-04-10 06:15:24.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3488513b2182dccc44e8cd7a3ba62cd275f75867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3488513b2182dccc44e8cd7a3ba62cd275f75867"}]},"branch":"refs/heads/master"},"d99a563e4db7424ccb50f2300e105c59f0cf2063":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-13 07:27:00.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/92/974292/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/5"}}},"commit":{"parents":[{"commit":"c5ad3cb5d3a04d5571161f0082397434908b6660","subject":"Doc: Documentation updates for ibm-svf-pbha feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5ad3cb5d3a04d5571161f0082397434908b6660"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-13 06:03:07.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d99a563e4db7424ccb50f2300e105c59f0cf2063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d99a563e4db7424ccb50f2300e105c59f0cf2063"}]},"branch":"refs/heads/master"},"06f2266fbd854bcf05fb5ec294176345f63adb29":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-14 11:04:51.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/92/974292/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/6"}}},"commit":{"parents":[{"commit":"becbd26a63eeccc83659a35cb0b2ed12731c7af1","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/becbd26a63eeccc83659a35cb0b2ed12731c7af1"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-14 11:01:46.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06f2266fbd854bcf05fb5ec294176345f63adb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06f2266fbd854bcf05fb5ec294176345f63adb29"}]},"branch":"refs/heads/master"},"af5bd663f49f8a69ae6912ebc2dcb061ccf08996":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-28 17:53:21.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/92/974292/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/7"}}},"commit":{"parents":[{"commit":"d3d529246cb447e69e26c7a87614d004b012a835","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3d529246cb447e69e26c7a87614d004b012a835"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-04-28 17:22:41.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af5bd663f49f8a69ae6912ebc2dcb061ccf08996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/af5bd663f49f8a69ae6912ebc2dcb061ccf08996"}]},"branch":"refs/heads/master"},"e6355f142fbb60d6045cad9cb960678d6f09f6b4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-30 15:08:44.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/92/974292/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/8"}}},"commit":{"parents":[{"commit":"df183ef5dd38a6992a50d5d870566d4c852dfec4","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df183ef5dd38a6992a50d5d870566d4c852dfec4"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-04-27 18:42:47.000000000","tz":0},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6355f142fbb60d6045cad9cb960678d6f09f6b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e6355f142fbb60d6045cad9cb960678d6f09f6b4"}]},"branch":"refs/heads/master"},"e0abbd3df52280da3a979420c5ee70120b89c78c":{"kind":"REWORK","_number":9,"created":"2026-06-08 10:00:10.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/92/974292/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/974292/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/974292/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/974292/9"}}},"commit":{"parents":[{"commit":"4b2cbdd9549c3e5ebcd09c612e954362321a6d49","subject":"[IBM SVf driver] Support PBHA group-snapshot and restore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4b2cbdd9549c3e5ebcd09c612e954362321a6d49"}]}],"author":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-01-22 06:02:03.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-06-08 09:59:41.000000000","tz":330},"subject":"[IBM SVf driver] Support PBHA clone-from-src","message":"[IBM SVf driver] Support PBHA clone-from-src\n\n[Storage Virtualize family] As part of PBHA implementation,\nadded support for create_cloned_volume from source volume\nand create_group_from_source from group-snapshot and source\ngroup.\n\nChange-Id: I4904571a460ad12b993214bc82aabdf087a600bf\nSigned-off-by: vdhakad \u003cvrinda.dhakad@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0abbd3df52280da3a979420c5ee70120b89c78c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0abbd3df52280da3a979420c5ee70120b89c78c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
