)]}'
{"id":"openstack%2Fcinder~975638","triplet_id":"openstack%2Fcinder~master~I0d3bf000998f7a0b93a76eefa37437d2a2451cbd","project":"openstack/cinder","branch":"master","topic":"ibm","attention_set":{},"removed_from_attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-03-03 20:00:38.000000000","reason":"Change was submitted"},"36800":{"account":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"last_update":"2026-02-13 09:47:07.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-20 16:57:06.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"38067":{"account":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"last_update":"2026-02-19 12:39:31.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-03-03 18:11:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"36439":{"account":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"last_update":"2026-03-03 20:00:38.000000000","reason":"Change was submitted"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2026-03-03 20:00:38.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-03-03 20:00:38.000000000","reason":"Change was submitted"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-02-20 16:31:03.000000000","reason":"\u003cGERRIT_ACCOUNT_37328\u003e replied on the change","reason_account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"}}},"hashtags":[],"change_id":"I0d3bf000998f7a0b93a76eefa37437d2a2451cbd","subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","status":"MERGED","created":"2026-02-04 14:42:18.000000000","updated":"2026-04-05 13:04:50.000000000","submitted":"2026-03-03 20:00:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975638-ibm","meta_rev_id":"970d94b8535f0c792b7f99581ee81dc7829fc041","_number":975638,"virtual_id_number":975638,"owner":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 20:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"all":[{"value":1,"date":"2026-02-23 08:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":2,"date":"2026-02-23 08:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-03-03 18:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-04-05 13:04:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-03-03 18:11:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 14:43:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-04 14:52:02.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-04 15:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-04 15:37:15.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-04 15:37:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-04 15:39:54.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-04 17:51:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-04 22:10:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-05 12:38:48.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-02-13 05:29:40.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-13 10:49:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-13 11:43:29.000000000","updated_by":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"reviewer":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"state":"CC"},{"updated":"2026-02-13 12:54:30.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-19 10:33:31.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-02-19 13:16:32.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-19 13:26:29.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-19 13:40:39.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-20 10:47:25.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2026-02-20 15:06:22.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-02-20 16:09:15.000000000","updated_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"CC"},{"updated":"2026-02-20 16:12:44.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"CC"},{"updated":"2026-02-21 04:32:00.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"REVIEWER"},{"updated":"2026-03-03 21:37:56.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-04-05 13:04:50.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"}],"messages":[{"id":"87206606c67a08be66556f1fe08d04987dc7a5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-04 14:42:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38750682dd618654f2e2805f8feee963e66ac52b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-04 14:43:06.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83ba96e43b9aa223c7bc147b26801351c58c90ab","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-04 14:43:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01f58dcd27a7b64d5855ccb2f65ad2a977efe978","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-04 14:52:02.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90924c069d6de6619f407c6a6cd4a0c4125d39c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 15:06:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d56a3dbdf441e4bfa9e797c5f48232\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8b710d3740c41f88af01ef77cc11432 : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c7aac53eb1142099248ab9d38a519a6 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ca23c109e51f4e61a4d300a3e26f11f6 : RETRY_LIMIT in 5m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4947c3b5db0ec3d7fadf1f3519f2cf886fb36f9","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-04 15:37:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/fc20e9b15f334d2283170f67dd37106e : FAILURE in 40m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bf063649983ac0e20af3941077d74c0ad33eae21","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-04 15:37:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":1},{"id":"b4b00e7240deff7f53b3001af1b32acb7dc75559","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-04 15:39:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 29m 59s","accounts_in_message":[],"_revision_number":1},{"id":"453c2b72b45e8295b92b07c04bc2f1e667e278fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 17:51:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f45054c993541eb901fbd8d123d7480\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7750560b3a454db1c286960df0a224 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1d768ba39654a7e906dbcbb357a2772 : SUCCESS in 13m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aee0da53780940958e6c64ba0eb1b485 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19eb78bfe424379a2789e2309cc9453 : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/170268bf3f0a4e0790609cb2e49e1112 : SUCCESS in 46m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cb0ffa186d147ed9296b9f0e7db46ec : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e7da2c45c23481d81d110cbfaef3a35 : SUCCESS in 1h 42m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88ca9d618bfd461b86e5e79c008a83d1 : SUCCESS in 1h 04m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26aa036e30474798be4b80331d0a19cc : SUCCESS in 17m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d1d1308009143cd854eb048f0f0a52e : SUCCESS in 19m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/756e0fe22cab4d47b8e5a88e0eeff287 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5cada505fd624d1db558ad538dac02ec : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b2c72e528014209b333d9845cc2f3a3 : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/22531500e03f468383a4ff2dfb1fad30 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/159b512230d045e88172b8afefcfe8b6 : FAILURE in 27m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecca1a765c8d4b31b8440ff7e6590f6c : SUCCESS in 3m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8bceae9e5c2647a88692172e1b29ded3 : SUCCESS in 1h 49m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8e24e9cedc2c40868ded05f8bfd0f48a : FAILURE in 2h 45m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88cb5918d9aa4855b6281eae990a9097 : FAILURE in 1h 31m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/40a30c3b842c47d8b7115cb7e3136992 : SUCCESS in 1h 43m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/13b7448495084043bbf457323b4a0ebb : SUCCESS in 40m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/21781fff072b465f9e79ab8394508f90 : SUCCESS in 39m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/19ed28dd7a2e45b0a20f7638a70583e9 : SUCCESS in 1h 52m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/942c246e0ae742c597f6318b19bc8912 : SUCCESS in 47m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/818cce47f81548728502611ad516afc7 : SUCCESS in 1h 51m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/d6ff2580fc594739b4fec86b2250943f : TIMED_OUT in 3h 07m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34b51cfb597c4e71b0badc0e509267e4 : SUCCESS in 1h 35m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97351322fe4f4ece950e3e5a65436021 : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0428acfa93f461d772900852bd036ae3886b9b29","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-04 19:06:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61be74100215579ac3d125ac1e9f10d5e6daaea7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 19:45:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1fb75b2d3484dc19e6dca5ea2bb9397\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75ede702170a4f02aadea97453b33e0a : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ce2f67a8f834b66aac1455c68a76100 : SUCCESS in 21m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f39009780daa4cc2a3de7bbecaaa3593 : RETRY_LIMIT in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ef5659b448c6dee40a9b3b2ffe74d8c3c1c53cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 22:09:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f43fdd592344bbb815d7011bebd6088\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62da2b6a169543ee8669c303dab85d43 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cf5a8a738cd42c1b95699d3014d9c56 : SUCCESS in 12m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78df3e8cb3ca46c7a22c6eccede7062e : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f5033abce24d37a6e9b89dbf5d57cc : SUCCESS in 15m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/e11c3250193247d086226a24c725d3ce : SUCCESS in 53m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40c1659ac38e4f018d3da3f8a7e4d6ea : SUCCESS in 1h 04m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c20736bfbe4341e294a4d4f3bbf25fb5 : SUCCESS in 1h 16m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33c618e98bfd4563b117b66bb9326026 : SUCCESS in 1h 09m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49f4607034474a8fa8c83f111c8febed : SUCCESS in 14m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cff0875591b14b7fbaf12b84a1fd2ad6 : SUCCESS in 17m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dc94b035024746dfa7966c9f46d4b0bb : SUCCESS in 2m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d68f41d4469e446eb95bb6e73cf22ec3 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5558869675e64f29a0ed4dc6e7118646 : SUCCESS in 3m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2f517d21798046a5a03c8dad62e0046a : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc01bc0b2cfe44be952c391ead9cf568 : FAILURE in 21m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/85644d6feebf4cfcb194a1dff3e05b73 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ed754c4e38b4ab1bba444121e38ebfb : SUCCESS in 1h 58m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/731431000c214c2b8e1bcf4e8ea17d28 : FAILURE in 2h 55m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a13531a73c13411c91caeb32a2b78f3f : SUCCESS in 1h 47m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/24be35ffb31748b5b97c4dd99449ead4 : SUCCESS in 1h 53m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b47f12adc0f64de680feb940a23d6389 : SUCCESS in 23m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de7e8a4f36224c7abc1da468f71b216b : SUCCESS in 1h 15m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58935b4e355d40318f0bffd617e8d363 : SUCCESS in 1h 38m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/87ab83d17cd24e889e76154a475b450c : SUCCESS in 33m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68382e83336f41f6ad88b4ef8757c278 : SUCCESS in 2h 15m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f7066d71716344779f6aaed917b65e30 : SUCCESS in 1h 06m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3dc27c6dde0a41c187a5879a5370b1fe : SUCCESS in 2h 04m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dae54a311ea4f26ac2ce630a131ea4f : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7cdbb030d4809ec2498c0b98f61e85b2e002c9e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-04 22:20:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_02_2026_22_13_05\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"9d50cba5650cdc8266b10abbcec3f69d2c0c4b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-05 06:38:56.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"12280934b8e118da5c9fb1a1f243f450cd987258","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 06:41:29.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e95520b7b432a8ef36beffa71168c6b9e016a2bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-05 06:52:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"73a022b5cf73b5f2211d0610781d48d3bd013e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-05 07:12:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"cd06a1e3c97a4bff54f017f8dd26c6ac8731a6bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:25:59.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9811aa823f65095a6cdcf11218d98fd26bf04116","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 07:39:04.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":2},{"id":"605186c91579a2d5fb19235a2770f793b5f55e23","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 07:57:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/da3a8bc0de744bd0a650e0fa016128c7 : FAILURE in 39m 04s","accounts_in_message":[],"_revision_number":4},{"id":"58923f4ffb4757243af34c1ba1d9145444a5246a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"date":"2026-02-05 09:44:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a9026c9ed6b2c01ba9de52664c8104e3b6c77a9a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-05 09:52:45.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"586fe674aa54e9cfd3c4380663ed4f1c1a4c91d6","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-05 10:31:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/823964550aa445a79c35a57f24601d44 : FAILURE in 41m 04s","accounts_in_message":[],"_revision_number":5},{"id":"5ce8e8c73ee2bbee980fe318af374f34d6002812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:32:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6aebdc864b742ccb16fad2be2291b53\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6c25da783bd42f392aa345fd639ec04 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac67430642bb46fca76f1b7570535e93 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1b0650c0231437d96ccd82acc7dd201 : RETRY_LIMIT in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6a74f8f54ded3406b902eb96a977cb283c6bc3c3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 10:41:19.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":3},{"id":"6868bacac05b18b3d7c6018b192741c2095661f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-05 11:56:05.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 40m 35s","accounts_in_message":[],"_revision_number":5},{"id":"159e3f4ab494b716ad7e54010c3d3f73d91b7c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 12:09:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36dc4aa36a9d48e6a290f273ee633ecd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2762c1367b4ed19fc874fd8fc90b60 : SUCCESS in 8m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/932958677866401589dc88635b676e1c : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1e25c5eb3d241038ccbe3c7fe659e73 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e62789979924030b9540a1ce1ba3b78 : SUCCESS in 15m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/b5ee28ec00c145269d291c50034adc3e : SUCCESS in 36m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/101cba75dae04be58ac4efff8913bc7a : SUCCESS in 35m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5360ddecd80d412785075d3752503bf2 : SUCCESS in 1h 38m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6264edfd1af14b9887a8191930f3d769 : SUCCESS in 1h 02m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ec29afe9c042ef99c3775463c3fc71 : SUCCESS in 10m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/12df3283080a4140a52ac70fd6cf12aa : FAILURE in 24m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ede0e3900931401892b8e10736ed7fda : SUCCESS in 5m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc9e9e8e76dc42a79c75d508bb04a6b4 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4bce302504ca4cd9a5aa23945912aac8 : SUCCESS in 7m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c95ac8508cd246388b2d25309c695cbd : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b62ed69052ce4f5892e5703d62e2f232 : FAILURE in 30m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/960ee41c89d243fe91abd2c4fcbb4da6 : SUCCESS in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd506a5bff9b4458a26002545c01323e : SUCCESS in 1h 12m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3b9fcf03f8d4c56b79f8b12787b6b7c : FAILURE in 21m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/441627ac44864f85bb904e4583c27fd4 : SUCCESS in 1h 37m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c14ee273d4b34446b3d283dc8ee80e4e : SUCCESS in 1h 52m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f5d10c285c8e41c182f35d093a98f3fb : SUCCESS in 30m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1d16a270903b4fabbd4a14113df5ecf7 : SUCCESS in 1h 13m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a80eb1ec18814ec48c2a87d802597b65 : SUCCESS in 1h 43m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/838993d8208b4fdf9383463592976b70 : SUCCESS in 45m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81290324a4284d449545286110ea3cff : SUCCESS in 2h 01m 50s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a94567a360a94f658a682a476453e5fd : SUCCESS in 2h 17m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/af4392507cfc4e4096c6ce957c8b0712 : SUCCESS in 55m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8735cfd4bcc480e9d874a555c4ecddf : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":5},{"id":"cef7ec702db7a83b9c29120043c0c5cceb3b2eed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-05 12:19:46.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_02_2026_12_13_00\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"a6cb594044f49de780afde5e0b886375cda2943e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 12:54:07.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":5},{"id":"76b7e5f40ca0a225a9a2a46338e345796dd0f591","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-05 12:55:23.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":4},{"id":"3526225a5fba874df3d1e00e7984575c0f4eb6d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-05 14:08:20.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b17b7fcc7a36fb38de8c338115e95e2a3a91ea9d","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-12 07:24:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"36995ae9ab6935d3be4f78d7c836a2b4dd9de753","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:29:40.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5a813d1037844c7cbe7314790153d788 : NODE_FAILURE Node request 200-0000623352 failed in 0s","accounts_in_message":[],"_revision_number":5},{"id":"2c68a69ceb3e21650a176e8f213d18534fb22dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-13 09:22:24.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5a489081c4aab189b4ba8465687faeea1034ade6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:47:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66e1d102c9254bbab3c89ecdd851f84a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a71bc2b16bc947b5913f0f2013d58513 : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1824765cf8494c8fae8550dd3e32df65 : FAILURE in 10m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05136f7a43574f6b9558e37e91c5b7cd : RETRY_LIMIT in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c234acff05778dac0a9ebdcfceee88fb190c978f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-13 10:01:03.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":6},{"id":"27cdb42b5ef15e9a21ba3e040cc001dc2a5906ce","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 10:06:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4d8ed98551dd4de998a319647083d2b9 : FAILURE in 24m 35s","accounts_in_message":[],"_revision_number":6},{"id":"ec8d378b091a1d79bc9a1a5588f1562f32a5d4a5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-13 10:25:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 35m 35s","accounts_in_message":[],"_revision_number":6},{"id":"ee16f708105ab2224fd282d6f43217e14477deb4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-13 10:49:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0d688a1cbb074955b11fa9ad3bb4dd38 : FAILURE in 40m 29s","accounts_in_message":[],"_revision_number":6},{"id":"195fd704428ddbbea898e71baab22f3986e8dea8","author":{"_account_id":37888,"name":"Akshay Ranade","display_name":"Akshay Ranade","email":"akshay.ranade@ibm.com","username":"rakshay"},"date":"2026-02-13 11:43:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a5395fc7ca48166f97643f68f6a07813ff37042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 12:09:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6f0346c21834cb5ba3f0831cf309265\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186d5a1727ca44b199eb1976d066134f : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e90d68f9c7a4991a464b909b10c4549 : SUCCESS in 11m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/281d293add894a0a8ff52508d405890b : FAILURE in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de558caf34c242d3ac5e02883707cf3e : FAILURE in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/babb8b6f42224524b03eddeb82a55223 : SUCCESS in 53m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5177989f13ac47cfab4f6bdfc5c4f7d1 : SUCCESS in 28m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11dd5d2b2c334814b73b8f1d54e8be55 : SUCCESS in 1h 28m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d98f6b8e8d924caa85b00034807e7dd1 : SUCCESS in 1h 05m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a742c8ad764142aa907c50466ffb3e : SUCCESS in 16m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a5edaf9532394d59aa0c73b7d783b5aa : FAILURE in 3m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd9cc4e55c0d42ef8b63b339d83fc5be : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d33393cf683c4859940e221d83075925 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a11964b71bf24a2dbea8b69157ca2848 : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d58a877350c04ee38f46fc6655de8ef5 : SUCCESS in 6m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7c7b2201186e4344aad3d79860d2bf42 : FAILURE in 48m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/87c3117d86cd4dc7821dd3f2ffe3a1c2 : SUCCESS in 3m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c3874280a194dbaa9f48b409d39f997 : SUCCESS in 58m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e25ae6907754f129c14c02ec45ccc35 : SUCCESS in 1h 31m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c3aba1f3a37a4953a831c7ff78cce3fc : SUCCESS in 2h 16m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/852041c157314d50b28fc038bac9b873 : SUCCESS in 2h 06m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e1e9e4f0b8244626894bb93884dfea5a : SUCCESS in 22m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/168e4e3c11bf417282cd41d39628d7dd : SUCCESS in 1h 11m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9d3a17b97814242b67682d1f8bfea3c : SUCCESS in 2h 09m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd66452d6a1d427a916af290ac9c651f : SUCCESS in 43m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/12f1b8f25aee429dad8965e89eff2bbc : SUCCESS in 2h 32m 59s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/94c64dcbbf214c11b746257685fdc430 : SUCCESS in 2h 34m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cef6ce8acb6b433fb661a080656a708a : SUCCESS in 1h 43m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a1e30da15854167adc938b977d999a1 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":6},{"id":"68cdd2041f4163701b438bf449dc952266829a65","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-13 12:54:30.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_636/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":6},{"id":"557530cecace627ee2e1805357c6d54387dc1b2f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-13 23:08:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62c1279631d14c5ca8ceaa09c32d3884 : FAILURE in 1h 05m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0fe3183fd2a743e899faf0b23f722202 : RETRY_LIMIT in 11m 09s","accounts_in_message":[],"_revision_number":6},{"id":"1b48e02c05bba36037907ca6dab87d4542cdad38","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-17 07:00:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"305567d622f5d5ba7be4edf518ea4040858ae1b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 07:20:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40600a4ba06a414b8db9e5af17b33ac8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a6fed29fdd141cd851acc211607a9eb : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f3ba45058d047c3abe96abb7e321e8f : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cfbb132db58444d7a03603726df8a652 : RETRY_LIMIT in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d9a2a014bec1d820689b35c1f30c931a9d12d81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:17:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af62831508cc46459fd2f185ddea2dd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed182d380c1d47309e4dec476d13e59c : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a04ebf911764a1eb1ab70c0cf68f936 : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c72ea97688941579df8d0a36ab72357 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a634170de7d4f5fb74e07e29dcbf7d2 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8a4c6eb646b24078b24d0f39d2d3db98 : SUCCESS in 1h 00m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aaaf87e818e44eb589d8277364868470 : SUCCESS in 45m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0be20d4539b48639dac877596537af1 : SUCCESS in 1h 28m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a55e0709deb045fb849f4ec1df0dffa1 : SUCCESS in 1h 04m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea4ae0c52a9c44e592679ef0b69bb4cf : SUCCESS in 17m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b8c994e8a77a4dc180f85027928602d3 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d224b85a3f1e4ba89de0ef85dc3afcea : SUCCESS in 2m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b60f2d499b04222af75b09bba1a00a6 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1508017f921f49dc9e826018d3bff3be : SUCCESS in 5m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8f8d6a433bd84d579182c1503fb1551f : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e31609457bd64c4fb0d6c053311c025d : FAILURE in 38m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/feaf475b3c554c7599069f397d1df4d5 : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14ea3b6d6786427080e24b4d40d16a22 : SUCCESS in 1h 22m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85d8a53d908c4e6e92b684126f0cd571 : FAILURE in 1h 24m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c961eb171b2e4eb188cc819c89f70c91 : SUCCESS in 2h 09m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e3cab114076e4133bdbeeb84e298fee0 : SUCCESS in 2h 07m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/91c526e53cc7451eadca449fe1a8760c : SUCCESS in 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4ea78524d5e4ea9af0151888682d01c : SUCCESS in 34m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e3412a7a3d0e473fae5dc382b5b1444f : SUCCESS in 1h 40m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/be9f60e412394f568da3675d185e647a : SUCCESS in 37m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6125d75748164f4fb0fa86e144e32711 : SUCCESS in 2h 00m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/1aee55899705448f96c6784d42fbff3a : SUCCESS in 2h 09m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a04f74e3b3543d0b136c69c1159d877 : SUCCESS in 1h 45m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/173cd0e1406049f4b5dd2eb45345775d : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":6},{"id":"2030165a91339873413499810befdcdc9302b01c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-17 09:27:09.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_02_2026_09_20_25\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"75f27061ba2b181eef947cd7728c6e8ce0cc968c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-17 15:07:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5b3bb3e34f0842ddb7b59e0fbc0bfd2b : FAILURE in 14m 37s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/2b11ea1cf46446818df0f9ff0754efe6 : FAILURE in 14m 35s","accounts_in_message":[],"_revision_number":6},{"id":"e956ccb173ae76a5a4d26aef044e573574ae9e09","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-18 13:22:55.000000000","message":"Patch Set 6: Code-Review-1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ff133104c8f08b77e2e06af1ad8cf44630346d21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 10:33:31.000000000","message":"Patch Set 6:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/797/","accounts_in_message":[],"_revision_number":6},{"id":"bc447f9f6cdea2897f19af496cd6d4e2851ba83f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-19 10:39:50.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/6/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/6/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":6},{"id":"6bfe05898bffa2f0b4f2596270a0629cc8747d3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2026-02-19 12:24:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"129fa9fdee3e9e68f0ade9ec44ee541707b7cab0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-19 12:39:31.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/53d677ea27074d4c9b807b34a9710df4 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c8fa1c95e3f64482b2a89a2e68c93c4d : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":7},{"id":"14bfc8d425d5dbe5540f9b7085869352f01ba009","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-02-19 13:02:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3525c394ecfd26cb244bc6bb93977ef4da515595","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-19 13:08:40.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":7},{"id":"4dee2579f2d1d4a767547cd3e487f5257ede5d3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:15:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408464262e8542a2bf5cc7574675336b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fcb6acb1b1f7405d8cea9b71ee42ecd6 : POST_FAILURE in 12m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/400f5cc81d854c3cb9b0ec36b2ac2b1f : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fdab549b2af946bea55b1698cf81d88e : RETRY_LIMIT in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bde67a4b7c824c657cd6f6d6a12190db9aec4d6a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-19 13:16:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/974080e38f3f4501bbcc1ed3837a43e8 : FAILURE in 33m 13s","accounts_in_message":[],"_revision_number":7},{"id":"dc8a3ba02d45cefd8a1d8d5967b6dea2be74068c","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-19 13:26:29.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"e0f3a833346885fb5c77e6a782390ce7339c56fe","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-19 13:40:39.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"8dca7b74fe9ea77a488bb314ef81cf28cfc0238d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-19 13:46:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 45m 02s","accounts_in_message":[],"_revision_number":7},{"id":"ecfbe19d0613bdfb1f10ee294969ed64b87532ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-19 13:47:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/169479c763ca400a8c1004cff91d7b95 : FAILURE in 43m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e670c7b14e5108633e3821afd668781318cf228b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-19 13:55:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/975394cedf7a46a6b0758849db33d96b : FAILURE in 45m 16s","accounts_in_message":[],"_revision_number":7},{"id":"ff6cd66b7a33b4630175df70ff4a2d48781a4e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:49:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/58d5594feb544c38ae2ee25f6bda0414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e6644de2a047de8c488952c14a153d : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeee04d32fc8457a914e5bf30fa29979 : POST_FAILURE in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2651b581133e4538b464413f58708238 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3380a00ad8a8475d87f553f78c5be4c4 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/242a0d3bd75c404b9bf92578fa2034b7 : POST_FAILURE in 31m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ecfc1ecfa074b75b5a6d3ec0ffe484f : POST_FAILURE in 32m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9714ac509f234a68a58f7892bb88bdd3 : SUCCESS in 1h 58m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8af1e887c2f34a659ae33132bcff6e96 : SUCCESS in 46m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b886c8f316f4190bc298f96d3aa3ef7 : SUCCESS in 9m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c8ec73575d4c46a3ae956644ff147ada : SUCCESS in 20m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf537812320c4f4ab1d31367c4d8660e : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/30953354578e41bd872ab922d77bbd5a : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33847d97effd4f998990b69b847e7a84 : POST_FAILURE in 4m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bd0cde4f1565435d855109ade3bef979 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0393af75124745738fb9d5f768906f4a : FAILURE in 39m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b3e1cf526a345cf904d35987b2155cf : FAILURE in 3m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d42e7314ddd4b3dbadad79a7f0a77f8 : SUCCESS in 1h 41m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9f06f3de1c564b0bbf4df9e4a5265f25 : FAILURE in 2h 19m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/207131259e3b4ab7ac49c36aaa40dd71 : SUCCESS in 1h 29m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/61222e784f1240b0b6a98d2fc10a735b : SUCCESS in 1h 32m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e669f06dddc14157b3d028147c4e2967 : SUCCESS in 23m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4de13f66cffa4a69a420a04537500cd5 : POST_FAILURE in 1h 09m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d053890a06504da28756f5517edadf99 : SUCCESS in 1h 49m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/93f8ed10e97b49a3a8d5f7ea23b147a6 : SUCCESS in 50m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1f35583dc7e74972ba03f251f17b988a : SUCCESS in 1h 55m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/72fed3e566054ade86501221fc02ab02 : SUCCESS in 2h 09m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17bfd9956412422e88879d856d7242ac : SUCCESS in 1h 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d51ff4cbeec44598ea6d662581937da : SUCCESS in 1h 04m 39s","accounts_in_message":[],"_revision_number":7},{"id":"e4d8d3a8f8a0f5bd8b759298bfdc98ca4c0d1220","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2026-02-20 10:47:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"db949d16d1206b51ca32aad9d7ef20eea488f14b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 10:57:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ba36364956364a63b9acf26e71644253 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ae3fdb9711374e268a2b99356c029eb1 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":7},{"id":"b1e47f3c4f7826de04b821a66b5bc6db202f0f8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 11:09:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c9c5c855c914e59b825cd2b21cdb2f5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7291847385e04987b16724c28bafcdad : POST_FAILURE in 13m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b4fa625de32b42e2920fd780716157a3 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a09f9d6bbbb84b1b853d13328b44ce72 : RETRY_LIMIT in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d9aacb890c7791dad3cd805eedf49028f38c40bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 13:21:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/968f1bc85f494829810fc99c898ca71a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeaac2cb702246f199285415ed6d1346 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b901d5c1789745f2b6acef133c40e75c : SUCCESS in 11m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c925548a5dd40048d6c1ef5bd0119b2 : SUCCESS in 11m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6573e42fb9f842eea5f800badaeb5f69 : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c212c5d59c40149f7ec45c3436d0fe : SUCCESS in 13m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/48a08832dc5149e29da86ab7a958dac0 : SUCCESS in 52m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54431ea0b0de446c8ea949d5d8c8f41c : SUCCESS in 59m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2806d671f0e545379c9e87e56d13b730 : SUCCESS in 1h 28m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c7521ecccda485fb267ca89ed14ae82 : SUCCESS in 1h 01m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bca3a7333d346af89b1673b79a14d56 : SUCCESS in 17m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/04cbf46d75774aba9efeec1988de564e : SUCCESS in 30m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d5abcac8fa964320a0a9c8835b739d6c : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e76ab929a864c2db10beb42ef1098ea : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ffd8ec7f78f44a1b268c290451b9983 : SUCCESS in 5m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7800cee4d3ee4ad39b2c6bb8310460c0 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/53c1dc9f5350467fb784b82f71bc9c85 : FAILURE in 32m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa2c0808939f4ef6aa66722ace81e5a7 : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3cacaad5169f46e0a2e8508fc3482d3c : SUCCESS in 1h 53m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1859789ac0642f18f40f4d57caef5d4 : SUCCESS in 2h 09m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a55c9b4e04774af18b3c3d6b40422d77 : SUCCESS in 2h 00m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/69090d99198749778d37f6bc1877d384 : SUCCESS in 1h 55m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0e8b19b10a48443a974c546429250621 : SUCCESS in 35m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9879358cf8544e58608cab6fe7c1100 : SUCCESS in 1h 13m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7273656806d3434cb8adee10d7c11d43 : SUCCESS in 1h 48m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca5e9381361645588d91728220f50ca0 : SUCCESS in 51m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/24938ccda69847c49c8054c6b9b4c6f9 : SUCCESS in 2h 28m 45s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2fdedd37f7c247faae4f7096faadf959 : SUCCESS in 2h 03m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/988b8407aa6d433b8a5459324a045633 : SUCCESS in 1h 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51744c64948f433cb6ad70b952505c30 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":7},{"id":"45e1afb41101d5f78c2336ee7212fdd010dc7b62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-20 13:55:01.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/20_02_2026_13_47_02\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"ea31db3d3bca165ea07b7666c41126db952d4ae3","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-20 15:06:22.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8bedb94d2ab621b61d6e6110b32a4560c09e4192","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2026-02-20 16:09:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"205a2309b952f7dd8450cdd7b386b6686775ce83","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-20 16:12:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b5049cc600c3424bb93db55f17974025cc851018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-20 16:30:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e08ecce6a5f7e2ff4b0acfcf89337bb13ce9ace1","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-20 16:31:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"72ddf5e013df9b63f7fdcc13f6aece1e6393b77f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-20 16:31:53.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/805/","accounts_in_message":[],"_revision_number":8},{"id":"04ee9e9b0d43275c0f5896e6e1bb4b4bc88c583f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-20 16:33:23.000000000","message":"Patch Set 8:\n\nBuild Aborted \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/8/fc.log : ABORTED","accounts_in_message":[],"_revision_number":8},{"id":"eb03518bbf3fbb6c0ca280d64b6ac14fecaae569","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-20 16:33:28.000000000","message":"Patch Set 8:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/806/","accounts_in_message":[],"_revision_number":8},{"id":"fe9252603647befdbc4d25033816ae0a0287a219","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-20 16:35:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e5143fee4b2a426aa5321502a33c4941 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b955bddb0dc54f67aa3afdbe4444e2fe : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":8},{"id":"8c9a3d486a5c935581bbdb6fd40da336af9c4e65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-02-20 16:40:22.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/8/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/975638/8/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":8},{"id":"19496d4de857e47d026c777a0b1ee9671d17b08c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:45:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f487bcb8a1684f99824aefc60f1bc8a1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dc58f75f74047a79135e95477f00182 : POST_FAILURE in 10m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f89e43c53cf415cba15744456a8a645 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2eacc6217924ca0b7c5384e92992a23 : RETRY_LIMIT in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1b7b4c8a65f6c3633d6af175f1c690a443d53e84","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-20 16:57:06.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50604d8301638406b7617fa49432dd415c0de49d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-20 17:07:03.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":8},{"id":"28b8e6a5d21a8012f7ba4d12c7851524d122aa34","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-20 17:57:37.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/d149aa94b37645c49140d014057d0a1e : FAILURE in 33m 15s","accounts_in_message":[],"_revision_number":8},{"id":"027455720f3b1961f2244c5c86799e99c845faae","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-20 18:20:29.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"1384d1ea903eee2baf73c0bf3d2c75a92d8b4bf1","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-20 18:29:20.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7984dff07ea84f84a6684093d9cd92c1 : FAILURE in 43m 41s","accounts_in_message":[],"_revision_number":8},{"id":"1d4f93baa2e3ffe75c714e1a03f50da5099a74ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 18:37:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fd603f5e50142fb8c79ddc48a0a2d03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f210cc69b74e0b8b976fcd6ab8bb32 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e4eff4a48944c648bbbf425161a974e : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e64418eba5894570ba49e4a691dd419d : SUCCESS in 9m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fca1603a0104b53b22cabc519de9b4f : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acadbd7990af4a88a6f968066979c58a : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/4c54b06cfc864944b69a4fe4d74b0fa7 : SUCCESS in 42m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc90ecf63a764627a5da4c0fec0011bb : SUCCESS in 35m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cd2d97d2b7f4c5b89f825492f8cb8ee : FAILURE in 1h 24m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/576a9e733c45435594913869bf4a4d1c : SUCCESS in 1h 09m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebfd3c3b8f134e659ee8c0ddcee4b5e0 : SUCCESS in 12m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f1161a4a5544ee9bfc6088255e56546 : SUCCESS in 20m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc1af62cfaf246f2812aa259f554c1c1 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7015b45ab4f84660a5fbb0f6b33e1fcf : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf63d95b647d4c9580023ccb1d99f5e7 : SUCCESS in 4m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b96e068f20d4ce182ac28016d7e9867 : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d6e9b03fc6b649078bec5204d380b7df : FAILURE in 44m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32ae911fbb5a423ca07f9e6310009371 : FAILURE in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5603b34fd0024c12b67fc59f7605d919 : SUCCESS in 1h 03m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f2c477b3fd6845b79a4b58dd1411554a : FAILURE in 1h 32m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3476c95a061744e1835ab13f808579b2 : SUCCESS in 1h 59m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/09955dff1acc4367b487095ff0318f13 : SUCCESS in 1h 52m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e6a22e29532436db5ec1fc1efef3974 : SUCCESS in 34m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/66be8da529994e8b8bc90aef047df544 : SUCCESS in 1h 08m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f308cbdd0514f3c9a7539dcc62ce442 : SUCCESS in 1h 21m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/27e3c37b6d164463b933940b0dd15b81 : SUCCESS in 23m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d98ceb1a56e1435ba9025356e6e85f35 : SUCCESS in 2h 01m 39s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6dfe902cb43c4453b8980d5faa0b25e3 : SUCCESS in 1h 37m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e358e535c2c415eab49cce6e1ffe45e : SUCCESS in 1h 47m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4d2f0597774452885e7b301232284bb : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":8},{"id":"3081963f10efc2b295dae1163cc76ae0bfef9e1e","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-20 18:43:56.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"4eb5f3fce08b57c53d33ac9f1b98f41b79ff7e47","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-20 19:02:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 39m 34s","accounts_in_message":[],"_revision_number":8},{"id":"5ae36565462aa341e2d4f6e7d92245b63fa13214","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-20 19:13:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/aa372594adb147ff9803a75ea2562ab7 : FAILURE in 40m 28s","accounts_in_message":[],"_revision_number":8},{"id":"1c98c8d3af80f9ad8d573e87a5a496cd56c9407a","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-21 04:32:00.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"91841ce325e649e4b1af84eb734bcdea7dee7a45","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-21 04:37:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5a2f2f951f234fecbb8e1cbed8a1f57f : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/58ec24200ebd4cdc8fc380a851465440 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":8},{"id":"806c63d1bd33a6c6b180deb867a714174712560f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 04:49:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4076aeebd3c0454181047320875bb2f5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13187e2345834d2f94909a9ec2a46ca4 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa6b3fa17bbd404ca07786cc7698b588 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bfd742c4408144b1a843f3571f283b40 : RETRY_LIMIT in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"86123acc13ae7e5d71636b9493f67fddd1d913ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 07:50:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd5b7d1603946b3b83489ddf865c230\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8084d725b8ee4f3ca7fa146297e36f1c : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94801d12034b4eafaec7998705062293 : SUCCESS in 11m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98f8bc5b5149433ea906caf07fd0af83 : SUCCESS in 12m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4dec7d8ebe0a41de8ab873c709442be8 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372668c187ef43fa9c77b9f2d17dda66 : SUCCESS in 6m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/d921056904fb4947bdd3683ed893d97b : SUCCESS in 56m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9f098075b0e404189e5c4b43bcf9d74 : SUCCESS in 58m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/253dd22b52c9460f8e31e7b0d760e023 : SUCCESS in 1h 31m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61be38b3aaa541e5aa851a3991d37e5c : SUCCESS in 46m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0c4be3b01c448aebf1f30a897767187 : SUCCESS in 15m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/383bc3e1fbd849778d8a317903deec86 : FAILURE in 24m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cf3fe311206549c086a01d2a4c2881ce : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22d30552f92c4d51a7dfd6a0df40fc3f : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2f8aac5b21941359d2d66719c09e73c : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bfca8a9f62fd404a9960c6de0c2efca6 : SUCCESS in 4m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7ca2c02c6f945dbb6529e6d9d066987 : FAILURE in 19m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/878fb0ec5c5a48abb11f5f7168aeaad3 : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/551335a6d5b84907b3a908ebcb332bab : SUCCESS in 1h 03m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/97251535ceb14a2e85a548134529c5b0 : SUCCESS in 2h 20m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c59570d6eb2244c498638a5528854946 : SUCCESS in 1h 59m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fbc7ea798cb9427aa8804f1aef13ab53 : FAILURE in 1h 45m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/151d7a6450874db4ad9565581af40c19 : SUCCESS in 35m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1f67981900f48d0859727703e718f67 : SUCCESS in 54m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d73831f8a5674916aaa1f8ad3d7d94be : SUCCESS in 1h 29m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7fcf80a4bc4b4567893df3131dd06210 : SUCCESS in 46m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd57e4632a78465bb98f2bf8bf4281ee : SUCCESS in 1h 36m 47s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/40d8a710dfb64bbcab6d9ebeeedebc9b : TIMED_OUT in 3h 13m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a7eafce59a04d9d86b1f03e2281e93c : SUCCESS in 1h 28m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f077e0345d0437b89dafc7215dce56c : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":8},{"id":"b97e72155aabbee1db4ff0d95c1e263de0cc69ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-21 08:02:00.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_02_2026_07_54_47\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"1f81b7a681f5214b070371932fde237c1a85bd10","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-22 05:13:55.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_800/index.html : SUCCESS in 87m","accounts_in_message":[],"_revision_number":8},{"id":"60205f34c1bb41999804f487a8a557cf297755df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2026-02-23 08:23:16.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"b4cd8464ba397b6409b53ba2fd91f52e3dd1434b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-23 08:36:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/42e33ca01d7341adb9e852b336100055 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/75dcf2c2f3ac43638ebea64bb295b54e : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":9},{"id":"3f3eacf02983235538e3a06b2ca7572bf5fb6747","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-23 09:01:36.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I0d3bf000998f7a0b93a76eefa37437d2a2451cbd .","accounts_in_message":[],"_revision_number":9},{"id":"1dcd679f8d63085dfd9ae0aa0473cf856dc5a3b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 09:02:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a91e07774741b58abcf89291159e7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02125b5196f445f49315d7a5447dcf6f : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ffed73947f5d43999f4a056c546df959 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fd32a4f7e6c4dc893e32ba4ceddf034 : RETRY_LIMIT in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"716a203923cea292ce0f6f2504611082562db558","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-23 09:06:30.000000000","message":"Patch Set 9:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"34f2f5797d18c094dd02334ce639cd39eec9fbcc","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-23 09:07:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/683308892e454c1a9bdaca9f83c873be : FAILURE in 37m 35s","accounts_in_message":[],"_revision_number":9},{"id":"ce14e1896591653e40ceeb8e46a59cceec82f1a2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-23 09:10:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/1a072e98bf9b47278591c71ddbcf36a3 : FAILURE in 44m 48s","accounts_in_message":[],"_revision_number":9},{"id":"164aef8b85d925ff7fd3a7436a1f1ec401896b01","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-23 09:12:08.000000000","message":"Patch Set 9:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975638 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"6c035b072b3a7cdab3312d9e3f50dd9a8ae06c39","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-23 09:17:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975638 : FAILURE in 50m 48s","accounts_in_message":[],"_revision_number":9},{"id":"9e03b90b78eea97a2615ce742c1f9ed8d83dc078","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-23 09:19:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4aaef6d80e054eba94b6060bde38d774 : FAILURE in 45m 43s","accounts_in_message":[],"_revision_number":9},{"id":"f808a53b191836bec3b1344a91f269041127c299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 10:55:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2622d6cbc1694772b3c1889e9f70323c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaeb1412d2c747bfb01e70a47438fb6a : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6e0cc2c8ad045d2933e72f713f9899d : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b06cb1bb2812424db8c67cf70b5237e2 : SUCCESS in 11m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60a1b1e128d54ba2a3761e7c1debef41 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717323697dd349cd9128740132d0ad23 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/7ad34a7b4e6e416f92a5cab7d9a33b31 : SUCCESS in 1h 10m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af4c665dff8e44ca89d0526857ce1c85 : SUCCESS in 35m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af4cd6abe22642d187647cb24934bd2f : SUCCESS in 1h 36m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2469dbd348d34df29589776ac94b27c3 : SUCCESS in 1h 03m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e68a769a23714f168879cf869cf6de26 : SUCCESS in 11m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1bcf2480da1e47a69ac8df6bb1fd9038 : SUCCESS in 29m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fbca448b66e34050afac11f3c4b083ab : SUCCESS in 3m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/022fb5cb384f4ef58981eff276e69e4f : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/927d07139af946c5b1fc351b58825416 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f1858d5f83224f5fbbcfef1df22b4cc3 : SUCCESS in 7m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b872057d975405e8c8b083fa5b5babf : FAILURE in 22m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c7fe67a96e6a468685a03886da7dbb8e : FAILURE in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39080008912e47bca361fc4a3ad6c5a4 : SUCCESS in 1h 04m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e0d42f9f9d54390b89e6dfdcfc0a4eb : FAILURE in 2h 15m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a4b7197046e4758b83144a0afbbc7ce : SUCCESS in 1h 38m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9e891bdd291742c68270743f8c1c89bc : SUCCESS in 1h 54m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9e4a593864d64943a96cca2c705de428 : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/807eb6e10020426bae24ecb1c8da56fe : SUCCESS in 1h 16m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0c18e9e87bc4b6581d0b47a12a42eee : SUCCESS in 1h 34m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a745096b9fb44f4a7aaa97dfc138298 : SUCCESS in 49m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b028c4cc60cf4ef6877a4f27afff43ea : SUCCESS in 1h 08m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/07402f39faa14b38be264739ffcf2684 : FAILURE in 1h 57m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9acf4ecbff144e289ca217759e78ad8e : SUCCESS in 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2e577034392470594547127c2f0f3c8 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":9},{"id":"0da8799704b33fd81e071895560ab4238e2b83bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-23 11:13:44.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_02_2026_11_06_38\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"691a361c081cb29b68c052c7b0890b9fb6c27ff9","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-23 14:43:49.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_813/index.html : SUCCESS in 95m","accounts_in_message":[],"_revision_number":9},{"id":"483f95e821d7d15b3160a26f77d2ea45dfaff62e","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-03 18:11:02.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f2709afd59fd0e61df0e6f2c3b417c80f0bd1deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:11:49.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0e38bcd3763bef95c60eecb2478665a80f052ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:00:38.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1440aafae8ff46219f4286a0df2eb305\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b37998ecd0490f91a4ce50ffed5509 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c3ad9d5e9345f8ad6c73e6f992cfd5 : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73c1399bfb8448fbb3b8e2cea632055b : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed4d86eae2054d5c8675bb6cafc881f3 : SUCCESS in 12m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/c3f48f4da07f491fb3f547ef6d38fb8b : SUCCESS in 54m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0c8abdca86947899167f38329b07ac2 : SUCCESS in 34m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7cc70c37f53e4fef8c9cdb60fb6f7d61 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/045631001eb04d1e8e5f591132457851 : SUCCESS in 1h 00m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5650b934d642439cb2368c4673bf99c3 : SUCCESS in 12m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55b3977fabbd4baab39e0fd876a23cf0 : SUCCESS in 40m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fbfee5f4acf045e4a268c0e1c6d16ed7 : SUCCESS in 1h 44m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/409a5ad4d42c48819867c8c6d9275cb6 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":9},{"id":"d1eb7f76459f76795bb63119eec1f890fe020e94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:00:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"43c490f2eec331b8ee99f381e499aea9221b3d8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:02:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85da8d841f02463e95dea5ca899408df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f092e49e742d445c91d6d91cba5763f1 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbc9b7ee8d3d47fd8d63a060afae6eae : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":9},{"id":"2c07a34830061f773f686833b0ef776aec2e5fea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-03 21:37:56.000000000","message":"Patch Set 9:\n\nBuild Failed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/933/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"970d94b8535f0c792b7f99581ee81dc7829fc041","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2026-04-05 13:04:50.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f96a21b5eecb4bef323884b54372c327c7712d16","revisions":{"fa535357897f50e9957777887648b6ad28b1d5c0":{"kind":"REWORK","_number":1,"created":"2026-02-04 14:42:18.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/38/975638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/1"}}},"commit":{"parents":[{"commit":"455d8710980a7ddc399f4eb579af55f6fd929f2d","subject":"Merge \"Add region_name support for Barbican key migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/455d8710980a7ddc399f4eb579af55f6fd929f2d"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-02-04 14:40:48.000000000","tz":330},"committer":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-02-04 14:41:54.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa535357897f50e9957777887648b6ad28b1d5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fa535357897f50e9957777887648b6ad28b1d5c0"}]},"branch":"refs/heads/master"},"c39ae69bc3a8ec3e66d080fee2d0ab91a53f6904":{"kind":"NO_CHANGE","_number":2,"created":"2026-02-05 06:38:56.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/38/975638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/2"}}},"commit":{"parents":[{"commit":"455d8710980a7ddc399f4eb579af55f6fd929f2d","subject":"Merge \"Add region_name support for Barbican key migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/455d8710980a7ddc399f4eb579af55f6fd929f2d"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2026-02-04 14:40:48.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-05 06:32:11.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c39ae69bc3a8ec3e66d080fee2d0ab91a53f6904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c39ae69bc3a8ec3e66d080fee2d0ab91a53f6904"}]},"branch":"refs/heads/master"},"8b1957cf2f4d509bfd6b211b2877bfec2b9a60db":{"kind":"REWORK","_number":3,"created":"2026-02-05 06:52:25.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/38/975638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/3"}}},"commit":{"parents":[{"commit":"b908de7ced66cbec32696b89eaec38739c5cdcc2","subject":"Merge \"Fix missing encryption params in Quobyte driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b908de7ced66cbec32696b89eaec38739c5cdcc2"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-05 06:49:45.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b1957cf2f4d509bfd6b211b2877bfec2b9a60db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8b1957cf2f4d509bfd6b211b2877bfec2b9a60db"}]},"branch":"refs/heads/master"},"05a6846fd46e7510e4703154e61479caba5fdea4":{"kind":"REWORK","_number":4,"created":"2026-02-05 07:12:57.000000000","uploader":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"ref":"refs/changes/38/975638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/4"}}},"commit":{"parents":[{"commit":"b908de7ced66cbec32696b89eaec38739c5cdcc2","subject":"Merge \"Fix missing encryption params in Quobyte driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b908de7ced66cbec32696b89eaec38739c5cdcc2"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"Harsh Ailani","email":"harsh.ailani@ibm.com","date":"2026-02-05 07:12:47.000000000","tz":0},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05a6846fd46e7510e4703154e61479caba5fdea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05a6846fd46e7510e4703154e61479caba5fdea4"}]},"branch":"refs/heads/master"},"68ea37c683f088d8886641eee8773e0b37965f84":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-05 09:44:17.000000000","uploader":{"_account_id":36800,"name":"Vrinda Dhakad","display_name":"vdhakad","email":"vrinda.dhakad@ibm.com","username":"vdhakad"},"ref":"refs/changes/38/975638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/5"}}},"commit":{"parents":[{"commit":"27af2c644dd9982162549a5c71c5c8ab458c4543","subject":"Merge \"Default disk_geometry option to unset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27af2c644dd9982162549a5c71c5c8ab458c4543"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"vdhakad","email":"vrinda.dhakad@ibm.com","date":"2026-02-05 09:34:13.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68ea37c683f088d8886641eee8773e0b37965f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68ea37c683f088d8886641eee8773e0b37965f84"}]},"branch":"refs/heads/master"},"0fcac00f704d5b4bdafa5cc48e8b3c86a8795d13":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-02-13 09:22:24.000000000","uploader":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"ref":"refs/changes/38/975638/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/6"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"sandip-rajbanshi-27","email":"sandip.rajbanshi@ibm.com","date":"2026-02-13 07:38:12.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fcac00f704d5b4bdafa5cc48e8b3c86a8795d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0fcac00f704d5b4bdafa5cc48e8b3c86a8795d13"}]},"branch":"refs/heads/master"},"66290fef3798eb89c4d4eb46d1d9b0c70e71f154":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-02-19 12:24:48.000000000","uploader":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"ref":"refs/changes/38/975638/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/7"}}},"commit":{"parents":[{"commit":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","subject":"Remove Windows drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"Dakshina Purohit","email":"Dakshina.Purohit@ibm.com","date":"2026-02-19 10:53:29.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66290fef3798eb89c4d4eb46d1d9b0c70e71f154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66290fef3798eb89c4d4eb46d1d9b0c70e71f154"}]},"branch":"refs/heads/master"},"7df1fc8a9d6442d5429423e2231725dcc68bfb3e":{"kind":"REWORK","_number":8,"created":"2026-02-20 16:30:21.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/38/975638/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/8"}}},"commit":{"parents":[{"commit":"2f43a2c058bbad3a385b258db46cd2b357f72eed","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f43a2c058bbad3a385b258db46cd2b357f72eed"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-20 16:30:04.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7df1fc8a9d6442d5429423e2231725dcc68bfb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7df1fc8a9d6442d5429423e2231725dcc68bfb3e"}]},"branch":"refs/heads/master"},"f96a21b5eecb4bef323884b54372c327c7712d16":{"kind":"NO_CHANGE","_number":9,"created":"2026-02-23 08:23:16.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/38/975638/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/38/975638/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/38/975638/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/38/975638/9"}}},"commit":{"parents":[{"commit":"2f43a2c058bbad3a385b258db46cd2b357f72eed","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2f43a2c058bbad3a385b258db46cd2b357f72eed"}]}],"author":{"name":"harshailani","email":"harsh.ailani@ibm.com","date":"2024-08-21 18:50:40.000000000","tz":330},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2026-02-23 06:00:00.000000000","tz":330},"subject":"[IBM SVf Driver] Added support for volumegroup-snapshot for temporary volumegroup","message":"[IBM SVf Driver] Added support for volumegroup-snapshot for\ntemporary volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for creating and deleting volumegroup-snapshot for\ntemporary volumegroup to allow user to volumegroup-snapshot\nwithout creating a backing volumegroup on SVC.\n\nImplements: blueprint ibm-svf-volumegroup\n\nChange-Id: I0d3bf000998f7a0b93a76eefa37437d2a2451cbd\nSigned-off-by: harshailani \u003charsh.ailani@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f96a21b5eecb4bef323884b54372c327c7712d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f96a21b5eecb4bef323884b54372c327c7712d16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
