)]}'
{"id":"openstack%2Fcinder~975880","triplet_id":"openstack%2Fcinder~master~I4b402d8ed270a1e1e1c702842c2c2004b8700392","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"33746":{"account":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"last_update":"2026-03-03 19:44:03.000000000","reason":"removed on reply"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"38614":{"account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"last_update":"2026-02-09 06:11:18.000000000","reason":"\u003cGERRIT_ACCOUNT_38623\u003e replied on the change","reason_account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-04-30 09:33:24.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-04-30 07:53:35.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I4b402d8ed270a1e1e1c702842c2c2004b8700392","subject":"Add support for AutomatedFailOver policy with consistent replication","status":"MERGED","created":"2026-02-06 05:14:01.000000000","updated":"2026-06-19 20:03:22.000000000","submitted":"2026-04-30 09:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":122,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"975880","meta_rev_id":"61647ee5fcdca5d103897cdd8ef7eaa619805dbe","_number":975880,"virtual_id_number":975880,"owner":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 09:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-04-17 14:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2026-04-30 07:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":1,"date":"2026-04-30 07:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 05:14:48.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-06 05:42:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"state":"REVIEWER"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-06 05:42:55.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-02-06 06:08:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-06 06:09:13.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2026-02-06 06:11:46.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-06 07:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-06 07:39:11.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"state":"REMOVED"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"CC"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"CC"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"REVIEWER"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-02-09 06:11:18.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-11 11:53:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-02-13 05:31:40.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-20 18:40:37.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-03-03 17:54:08.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-03 18:01:05.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-03 18:01:23.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-03 18:10:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-10 08:58:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-10 12:41:34.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-12 06:43:34.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2026-03-25 17:13:01.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-03-26 17:03:35.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2026-04-17 14:55:07.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-06-19 19:57:19.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"676b1c265472506d3b3ae361d92c497a08a79763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-06 05:14:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc8cdd9c3454accf067389881687254cedc65260","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-06 05:14:48.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cda433388db6cbf51e9072465498700c631dcce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-06 05:23:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d488a3704a2913a46a66b2d4ecb6848425e73177","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-06 05:24:02.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"29c76ef1eb1447b4432326f2f1b6c7915bf0c2c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-06 05:24:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd8cf1f1ac2f0fb86e357735b679ee527988acf9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-06 05:25:45.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a08720206282f25b710adb5f766874f522f881b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 05:42:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a41b25e5214e4761859ec7d416619f6e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b8f23c55b45471aacbd60486fee7176 : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8cc0fd73046d4dc0a3540d630dca91bd : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a93f540a2bbb4f79be3f5f5092814509 : RETRY_LIMIT in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b5f4e8529e9bb60db36ce9e5c0072ecb3ec299b","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-06 05:42:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e3b423341ee959d0f90d7bf6c548c6e8ae9b9a3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-06 06:08:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 40m 53s","accounts_in_message":[],"_revision_number":3},{"id":"975ac39ef0b40ec4d3c01eb92ad6306e6235a290","tag":"autogenerated:zuul:check","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-06 06:09:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bcf90a01a811446e92163b6e4b50a63e : FAILURE in 40m 12s","accounts_in_message":[],"_revision_number":3},{"id":"3ce82a0dba53b203976e0c3da1926b6c7d32145a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-06 06:11:46.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":1},{"id":"24a3aca20b362b15d515940909289be3df519c9d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-06 06:20:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":2},{"id":"b3607c6becf049ac49d4a378e6785f39149e711d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-06 06:24:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":3},{"id":"f4c99756f79a858df1a04f60fbeb618789891c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 07:39:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0b72129be646b0bd90882370a000a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf5611f9c4c14ac4b4ebcd00786b243a : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce55dbe9a3be42a78eaa8da3268357d9 : SUCCESS in 11m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa257a38496845d5a2ee9308ca941762 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbb733e261349c3a58c96829b5b9756 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d9fca51d8cb24778a4ef76b685947cbb : SUCCESS in 29m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e18466dc72140a98300941ef2f6ea30 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6cf7840c6654cd2a88e4e4bcf2eccdf : SUCCESS in 1h 27m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46d464df04e946ba8e5d84c2372d0c12 : SUCCESS in 1h 06m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f74d486c5e74a019eb086cdf09879e2 : SUCCESS in 9m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fcba23ef3a41459b8e236777005fa0e1 : TIMED_OUT in 40m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c560768a4a94dfea188d428cad1632e : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81601ecb972840948f3c522e64aabce4 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60b18f4fb4204a65b890e6578267ee81 : SUCCESS in 5m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b2d426b439644246866a925d87b7b546 : SUCCESS in 3m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/55909c1ec13d4587bddcc9695d29a9f1 : FAILURE in 26m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9ba292b19b04c868a41268527385839 : SUCCESS in 3m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b81a6e5590c4f7db0a4a84cb1c83109 : SUCCESS in 1h 02m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f78dd04b6ed4917bdc535d0e833a457 : FAILURE in 2h 13m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/486309f2b8c74bd39c87930309a1a360 : SUCCESS in 1h 55m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ec053999770945a0985af0eb732b64f2 : SUCCESS in 1h 52m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d0c81129de914732bb0c62bd33e4a19a : SUCCESS in 24m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/82c9720568894b49a1d932af51c77b9c : SUCCESS in 1h 08m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1ea16962e4254d97ab1305d4f42b16f0 : SUCCESS in 1h 10m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3f895b7a889b4e9dbddd9b8e5f37ad34 : SUCCESS in 32m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fe29120772da4154941269e6c5ddb045 : SUCCESS in 2h 04m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bcc8a694c1594428a975c2c463c50ab2 : SUCCESS in 2h 06m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/755bb941f19d450b842f8ff40079c0a0 : SUCCESS in 2h 00m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb87a9ff9ca041b2a0ac6f58ce14c70c : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":3},{"id":"4000f2bf889a2e27066d0dd09c0648171b5ff4cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-06 07:49:13.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/06_02_2026_07_42_18\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"f47b046337d0301d76d04731977692b9993113c6","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-09 06:11:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bfbfa21edd3d21b29c347b490e7ce30af7c9269d","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-02-11 09:38:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d08630828ad3f98acfc527c4af2434eb491b2e8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-02-11 11:53:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/5994e60/ : FAILURE in 2h 14m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/b7c59f7/ : FAILURE in 1h 34m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d028817b9d17c7afdf4ac7e8f75f62aea16fb819","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:31:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/35d0391d5f6f40f4bf936748e7350724 : NODE_FAILURE Node request 200-0000623371 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"31835d6244d05c9dcf5b4001feac29f6f7d123cd","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-02-17 11:24:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15b609edf5555c36fdc823fbe7c86ff96ac61c72","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-02-19 05:53:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1105c33ebf183fd32285296f5b2d8a8abb06ab1f","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-20 18:40:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e970eb79c017a7b65e9d23ca368e3a9b4ec506f7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-20 18:51:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1ef5b4491aa3c62ece15c45e1b9765c39cdf0dff","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-02-20 19:25:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fea98d488920ffb1d4f5a61b3e7531bdbda1e583","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-02-24 21:21:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a975c222a14e8d6dbf3c30642bdb72939d9131d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-03 17:50:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"651a377ffa0218e7bd92af7a891b5dcc96b5dbc4","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-03 17:54:08.000000000","message":"Patch Set 4:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"93ea036c8c1a9da28a8f57b35c04139117330ccb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-03 17:55:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/69133003f32345bdab4e349ae18fb215 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/035222c7b59641538bed94628a5b4ada : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":4},{"id":"8dd8a3d655593911b708f2a4a324d53123eb5ea5","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-03 18:01:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/f6d85d44d48f42de9b4058a887dd3249 : FAILURE in 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"74fbf8cf1ffe2c1cc984ae998a2c7114563a348f","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-03 18:01:23.000000000","message":"Patch Set 4:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"87b22cabd7c1d92fef4e9bf0bdeeedd566db7233","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-03 18:02:12.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c5009e469f8e4c1492b867dbc9dfd464 : FAILURE in 11m 05s","accounts_in_message":[],"_revision_number":4},{"id":"93b5864f1c58fef474c2402259ff038eb633e272","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-03 18:10:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/40af7ebba4cd40e8b84cb3ac62b73cfa : FAILURE in 19m 02s","accounts_in_message":[],"_revision_number":4},{"id":"dcad51117780bd1941aef63148062658488760df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:18:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39c0a84a5c54191be8343eef6588216\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/782c500c6b3f45c98082f4e27989d1ea : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d53389c348544ed2b412c9f2d8ea157b : RETRY_LIMIT in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/78bc1ab75e524c47b957948aee769574 : SUCCESS in 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"44f0586ed74f66abaeb4c08ea4f97963949a39e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-03 18:29:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 38m 01s","accounts_in_message":[],"_revision_number":4},{"id":"b342da219527e45e50fa339e506f2b35ff6810d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-03 18:29:53.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":4},{"id":"78c5699c7751d70b61eca40420a71fc1076814ed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 19:44:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3308/.","accounts_in_message":[],"_revision_number":4},{"id":"d781f367663fc0674bb0f55a78f02de7027719c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:09:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3ccd196fb045f395f641dc424351ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b355eb9a294770be2bedf7dbc93065 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71fe1b8b5734484989f3cdabc08a21d2 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f60b548714a47c4a1f2870fc62dac5b : SUCCESS in 11m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c5730e8ce314c82932366f3d18ba136 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c07303a6383414cbc896df6bbdc5066 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/18335abd5069400bac2ebba9fab7961a : SUCCESS in 1h 01m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe05106291bf4b7fb97676ae0ddad0c4 : SUCCESS in 53m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d03885224031430586258adefe8a054d : SUCCESS in 1h 50m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef5baf8eb87446dc81e60dbfa6b906cb : SUCCESS in 52m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634a854c5efe49dca931b96fd6abb6a1 : SUCCESS in 11m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4885763b538c44a19b779d8e2ba7d1c6 : SUCCESS in 22m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c5e35dd0d4d74109bc0288c982e69fc2 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78d0e85b909345618dbdef66bf12a559 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c73ad7d81cfb4c30aefde0ccfb4ee909 : SUCCESS in 4m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/70de17c44fe140ea985df17dd0b8abe2 : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4744b524a1b4c839620a4a84b242fd9 : FAILURE in 39m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ea3b161538047fcb3d08394fe55ad26 : FAILURE in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f648bc8748344d328d28b6963fdf7376 : SUCCESS in 1h 51m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a36fc3e9e44a42d4ba666aeafd0bbb3d : FAILURE in 1h 26m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb928453ead54aac8a46f7a0071980bd : SUCCESS in 1h 56m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/88a04c341ff24782afd07311dbf873cc : SUCCESS in 2h 10m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/30b7cf4042444215ac488b99555ce421 : SUCCESS in 36m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/24526733f2814a218f6e326c886e31e0 : SUCCESS in 36m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9627937d71b4ff6ab64585805d2bab5 : SUCCESS in 1h 25m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b29c085c7203427e8598692bc0d072ac : SUCCESS in 51m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/342beba9e7984c1e83e3dd66df53f8bb : SUCCESS in 1h 23m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/046d691ba2ee48d7a945b1bf036bea4e : SUCCESS in 1h 10m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/64ef4d2031aa40c5983df7afd2c15b7d : SUCCESS in 1h 47m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8c83914f3434ad78f03b62d7e534386 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":4},{"id":"375436cd60a5c7c1125ac28add24b076c9f9e8b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 20:11:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3135/.","accounts_in_message":[],"_revision_number":4},{"id":"165e3e47799d63eda905ec2482bd4a13273870e1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-04 05:03:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26e862bdc36be5d29b39263c19b2600eed49bb31","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-06 07:29:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4defbd1c2b7fe8e199b37780e5bc2d34f776940","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-06 10:04:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/4f68d86/ : FAILURE in 1h 52m 41s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/337fc37/ : FAILURE in 2h 31m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"f63e29c761afd4746cf45792e06a4311232178fd","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-03-09 06:21:35.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a645537406e07dfdc2acd6b8eae38845698bad3","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-09 07:46:35.000000000","message":"Patch Set 4:\n\n(15 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0041635e79a23845839fb8a883404b07316f9a96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-10 04:34:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0102d11d944f0cef243f64c5d0a6a3dbb2a11408","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 04:40:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/178766de08b04a85a0d13b0f2cc02571 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1ce456294c4748f0a66d6efefea6f553 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":5},{"id":"689c50c37fade5e256414d2b908cba29b095f7eb","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 04:47:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/f7ffde3fc44846edba9ae904b4b0d492 : FAILURE in 12m 07s","accounts_in_message":[],"_revision_number":5},{"id":"9eeff38d9cebb717991bc35119894a804eb762e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 04:49:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7783230e89d4e65b4a9797d6e85b6ed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/161eee8cdd404c91bfb1fa88a5630bd6 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c92f924800d94935a2b0bbb793282646 : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/46b66ac59419455aa0d29eb42b67f553 : SUCCESS in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"07d9cfe4bff063570e66618d40c55d01f4a27f5d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-10 05:06:51.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/5ce0acea78f642a881fb7d4110a1a633 : FAILURE in 31m 26s","accounts_in_message":[],"_revision_number":5},{"id":"fa6197d33d47688d2a6b885af5cbe2c59f61fa11","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 05:11:34.000000000","message":"Patch Set 5:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975880 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"40da6e5419c2366390ff65e7f98b45df7ae189f3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 05:12:07.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":5},{"id":"c778ca92785d073d393578287224408d8637a604","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-10 05:13:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/276103109b70406d9cc811a42e92dd30 : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":5},{"id":"4960ab600e7dab9f4da63c5f4d0a2addfe0ec147","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-10 05:14:42.000000000","message":"Patch Set 5:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"a8f8dbbf420929268011b4a01cb9e13f0b890d17","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-10 05:15:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 39m 28s","accounts_in_message":[],"_revision_number":5},{"id":"e24132b68c0df1e1a75c65b32710e65d8a310d83","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 06:05:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3409/.","accounts_in_message":[],"_revision_number":5},{"id":"86e3cbeb507e766052a57a56982ad8fc04ccc047","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 06:30:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3236/.","accounts_in_message":[],"_revision_number":5},{"id":"6e5a628b97b7400015d66c9caf6f9a20a0f81ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-10 06:54:35.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"36d4b130f62d264e20d7ab5a6564c19e9eb48b3d","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-10 06:58:14.000000000","message":"Patch Set 6:\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"50e096453155e7265125cd5eb0a49beae208f210","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-10 07:00:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1a794186e1eb4a5e96c611ad7481cac6 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/14d1224f4de14833abb4f4b03401940d : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":6},{"id":"f9c2a39b6ec315dfb60f5c712e2fa38dd319dfd6","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-10 07:08:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/9b1331fc35064958b9e141cef4eb9c9f : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":6},{"id":"46272ba0c48fc9af9415653961bdab358533fd82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 07:08:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a716932ec5043d9815a29c74d1d0a41\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b33104fe7e824131b8bf48cff78c4c19 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d88ebf1985bf45b1874b7caa08bff23c : FAILURE in 12m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bc1ca9abe195437fbd06f65f330c5cd4 : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"57fedda76eb5c06d35033fa1093da8c24c88bd93","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-10 07:27:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-fc-driver build/789d57be7e5b436c854e060ff70ad303 : FAILURE in 32m 49s","accounts_in_message":[],"_revision_number":6},{"id":"4b785c6c8e0f962c638419cdb345e5838a29eea2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-10 07:31:22.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":6},{"id":"dcb0544da9456063bb7e5dd5f6a1495ccfeed622","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-10 07:33:42.000000000","message":"Patch Set 6:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"a40e3f76672c0eb9f6cf6c8f64ec131faca84b62","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-10 07:33:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":6},{"id":"5b15cbc3a99d90190e31954dbdd5ad46352779da","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-10 07:34:16.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/353d8a7f237b4e33bdda52376d6511f7 : FAILURE in 39m 08s","accounts_in_message":[],"_revision_number":6},{"id":"53266f1ef5eceb397ccf83af3d6162fe4509005b","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-10 07:35:36.000000000","message":"Patch Set 6:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ee61600eecc6dd5ddd02681f2cbfa402dbb5e895","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 08:25:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3411/.","accounts_in_message":[],"_revision_number":6},{"id":"2c26241ac5d9d5259aff493cbc1e8e53d6425c3d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-10 08:58:17.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-6/FC/ : UNSTABLE in 1h 27m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-6/iSCSI/ : FAILURE in 1h 27m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"28647c464ca55aa8245c2f0e720488043ecd9a0a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-10 08:58:45.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-5/FC/ : UNSTABLE in 1h 3m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-5/iSCSI/ : FAILURE in 2h 23m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"4e99e6144ed30eb13fd66e89b6ea7cc32e2eb673","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-10 09:11:54.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3238/.","accounts_in_message":[],"_revision_number":6},{"id":"a06c197234a59b7fa9e8ed29be51ec9e574f167d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 10:36:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dc2993bc8af46839920ec0047918e4b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeeb0e77595e40b1a8893819bef89828 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cf6ee057e504549af38f29c6463db7b : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e12d5440e06841d6b537fe272351fdd5 : SUCCESS in 10m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1f506d547444ae58ba1cacb4156f98f : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3923e3a9cb7a474596b2ac3ac648374f : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/6335ab1daee24b44a90f0eda332c4f4a : SUCCESS in 49m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2fbc586627d740c89477120090d4ed7f : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/795ea6d5d9044f2dbcff70477d449e74 : FAILURE in 1h 02m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5f7965586334d518f14a7e4ca3ae754 : SUCCESS in 37m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3f29b9b57844f0bab174ba2551ba12 : SUCCESS in 15m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3af34c85674548a8a08749f9500cb30d : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc50cbec66ff42d78db34f102983674f : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3a8c382c3954dfaa6fab93b4cef0c14 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d3a6cc542ef49fea06b047a268cba82 : SUCCESS in 7m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01f913dc5006455e8ce8a04e56922aa2 : SUCCESS in 4m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed87ab2fb5cb4c05b038e8b9cc219b6c : FAILURE in 26m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f60f53c7ab44543ace5ed60600d482f : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52945f359dd84f1ab4c64bbca3bbd101 : SUCCESS in 1h 44m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b17b8447bc964498b40bf5ddb3da1d72 : SUCCESS in 2h 30m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9b21dfefe8e04ba5a7e2b5bc6d1aaeea : SUCCESS in 2h 00m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/510dc2af6f7d45409e8ffc9c666a1cc7 : SUCCESS in 2h 02m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1544a25648f144a8bead301dcdb8ac04 : SUCCESS in 25m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2d51d5c4b354d2fbe7127cd2cf25648 : SUCCESS in 1h 06m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b627b0a9f3214908b9752b4e1f48dbc0 : SUCCESS in 1h 49m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/622233ffaadc480bb403a9648164d0a0 : SUCCESS in 27m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4ca16ff1b5de408da675a2af45deb256 : TIMED_OUT in 3h 34m 43s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/04c63bcdf0814f9dbcbc919c22af567e : SUCCESS in 2h 20m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f89bc46282184a448e38ae324f0362cf : SUCCESS in 1h 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c863b4dd6bdd48ccb9d6e1f66e870d48 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":6},{"id":"d7b78584dd5059098786b1f4e6dfd9cd7f90b4d8","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-10 12:41:34.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1035/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":5},{"id":"2df20a568e780280d9e5cbc01d49e05954e1ab48","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-10 15:40:42.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1037/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":6},{"id":"6825a07eb0a6632503b9da9622302d5fbfd8926e","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-11 06:14:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9005d4f58081cc6c7022dd59dc19e7f0e39e1b22","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-11 06:20:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1ac72eb9118a4cf599caa88d019c83c8 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/43c3fa594d4a40639a5d551c22f0e002 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":6},{"id":"3d651ffe477e4e902aac633670cc9c1e0f5f4a5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 06:32:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c100432c1742b289dc129ab4f3c145\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fe740c73e384ebe89e09ac588219e01 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f50e7031537480e98c300f6c93e45df : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e370dbebcde74993a2a0aa1f39adb1ea : SUCCESS in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"17c4427e5b350666ef9a30d0e8c821101e0bfe9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 09:23:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8798d2829f3407a94bbbb839d420aa7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7130d701df46939ae237eaeea74312 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baa7e88ae254484cba2d715d820bd5c7 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9ea9e28d6ac40aaadf7280e8bad9f23 : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/301fea66066442c2ba104b45a0486ae9 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adabe6115e1d492fb0a2efdc66cec43e : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/ccbe4f10aa834094ba60459f91a4efcd : SUCCESS in 59m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f297af878844fd78f1744c5ba84ef77 : SUCCESS in 59m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02bcba3e9ad04555864f1d3993166157 : SUCCESS in 1h 26m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c7f4f48148443b787c1ba990a6cf233 : SUCCESS in 1h 00m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8149bacd4a345dca1d4647df5766255 : SUCCESS in 17m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d54a8ba6e7d45e199c4f3275641c8d4 : SUCCESS in 23m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/44524d832c7e4d0496d793ddf2d39790 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/830587b7b4384714b5d9768a200a046d : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea0c7636d87f4eddb33c3410453a2570 : SUCCESS in 4m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5e9a089e90e8414bb1d5df4f049f7eb7 : SUCCESS in 5m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee5418bb22954a108cd59dcab591269e : FAILURE in 19m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/88462af4df2344349cd9683d58662b2c : FAILURE in 3m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fec06c7d79e94f2e9f15cb51f6029a27 : SUCCESS in 1h 02m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fe987fd83e3143228a37b6d0b538e9a0 : FAILURE in 1h 58m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/946cc257b90648e5a535282433e98046 : SUCCESS in 2h 07m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dc0664ec3d3648d88946d10d8f62ffb3 : FAILURE in 1h 28m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/796d093eb0f94607957819f227929b3e : SUCCESS in 37m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff152d33996844448ce176cfa7aebdd4 : SUCCESS in 1h 13m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/82f7e933d82f4c16ac0adf1cd5f5bde2 : SUCCESS in 1h 52m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c7042e8928fc448aa366dbc747a29c97 : SUCCESS in 41m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0253b2c9a8e64be48f7a5686b0f9ecfa : SUCCESS in 2h 36m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8dda2c99fa2a41fe8d8f3db038864282 : TIMED_OUT in 3h 07m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/567a49bc02df4a538eb81654d4d15d12 : SUCCESS in 1h 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af8edd0286334c7c88c7496352a76aae : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":6},{"id":"7156e7b6fac37a1127e648d1ec26c16472cb2aac","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-11 14:49:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"92db845a80081ee0bc4ef2a8c19acdcf0516ed59","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-11 15:24:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/72f65f6/ : SUCCESS in 29m 07s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/54019eb/ : SUCCESS in 31m 50s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"189908e5905eae996523b048fabb0ddb443b01bd","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-12 06:43:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9b953eaae90dc2ce5b937279a7b79738ce7afe6","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-13 00:59:10.000000000","message":"Patch Set 6: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e3c2c061cd024844fe220bf4e3dd032415ff1aa5","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-13 01:00:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5fc3fbb30c61451d2d156026ca59cace836e6a1f","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-17 11:37:54.000000000","message":"Patch Set 6:\n\n(14 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4c11a10363242583c71fd049c503cd38ff25042a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-18 10:51:07.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1bc8b66aa0596cc793dd9fd760f600423e4c962f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-18 10:52:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e6bf03bc0578410f855e7fb35971bf18 : NODE_FAILURE Node request 200-0000654314 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"200c92482a64c991c65dc41c00d459b6d8534da4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-18 10:56:58.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/645b1ee01a7b43e28f093efe6302f8ce : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/17d45d6464cb48ed9c6aa46a08bea4e7 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":7},{"id":"6a93dd286fb78586fd572ccd69abe39259d183fe","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-18 11:08:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/515b182cf7924c40a71fe8027e270929 : FAILURE in 16m 29s","accounts_in_message":[],"_revision_number":7},{"id":"eb466c4f81ea7d14027f545871d3727183a4aef8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 11:26:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab92b1b066743bab81a537ebd76c71b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4dddd292a2a949c5bac3473857e7e648 : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cac1943d35e465facc53d37a4f2a5fe : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/66a2bb9802ae465cafac497f31d3d283 : SUCCESS in 25m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff67f58fc0b211ac455948c82bd7ae357ace18f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-18 11:29:47.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":7},{"id":"a809ca07486826cc4df1258d50919c120ae636bd","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-18 11:57:16.000000000","message":"Patch Set 7:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"935e7347097d1e1b86e0dd609c66cff2e8dfb519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:27:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7e4f9fd5e664c2ca3c9367f447b50b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e258edae296e4b6caddd5608be833bb9 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbbb9894bb934ca6bded2101d29a18f9 : SUCCESS in 13m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2aebc9a2aa894d33a036ac766b2688c8 : SUCCESS in 10m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/643561be4b27434b989ade9052224687 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46c64c911b2455489924dc633d6385e : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9c3ac961baea41f3991ee06afd58cd94 : SUCCESS in 1h 00m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e182a9e75a77409cbc3ce2a45c6304af : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dca7d688d0084620909bc6e6e7c5e3e3 : SUCCESS in 1h 25m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e83f12d8736043af8c4c9a3eb63116dc : SUCCESS in 1h 06m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d02fde6e6ad84824b7b1fa0b7d65e7ec : SUCCESS in 28m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/627361fc619f48639819cc50dd0c1395 : TIMED_OUT in 41m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e899af2b3f64073a56b7156ab0a6028 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a4f1fef68ac24dc6b386133ca1557afd : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfca0d9d65714fa0823f2418b96eb532 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7da7d67928994dc7a21caddcc786a474 : SUCCESS in 6m 33s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fc84bed539944a9b938a08a1239344ab : FAILURE in 42m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ac2e178ea09451796ceb9baa54e5934 : FAILURE in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ebe24803fba8445bb0ec3170dbb2233b : SUCCESS in 2h 02m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/22b21519b6a042119cb8570f274cabd8 : FAILURE in 2h 21m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c8756f4da71e410fa5d5e2d1a63ae83c : SUCCESS in 2h 14m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/183e382010c54781a5485ddcf111ab6e : SUCCESS in 2h 18m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/de4fde61d8674b71b8938c7bd1f0a8fe : SUCCESS in 40m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90e8c2258d9645eb94490a3bf8c7172e : SUCCESS in 1h 16m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17bcfa71dae2447cbfdbb0fbda17b755 : SUCCESS in 1h 54m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/635165603a0242528a2dfb04d11528ca : SUCCESS in 35m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc1c3cf647a445a4b4927148784b9bce : FAILURE in 1h 59m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ff3ba2ab2f9c4e53bc7e4ccad90eee78 : SUCCESS in 2h 13m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae3ed4d19b754ec59ec5578a9c1153b9 : SUCCESS in 2h 08m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01cef571df79468c950c1e7846fada58 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":7},{"id":"c7193438a44e1a2b6420b897fb7b6eecf7078450","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-18 13:27:36.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 39m 23s","accounts_in_message":[],"_revision_number":7},{"id":"5f0c5c78293ce1bcf82b094aed0dfd9bc23e6783","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-18 13:45:11.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c7c99878bcba4e788d7e6dad527639c8 : FAILURE in 38m 30s","accounts_in_message":[],"_revision_number":7},{"id":"c1a90079cd6cbec4c759978fc9a09663276bb205","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-18 15:37:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3522/.","accounts_in_message":[],"_revision_number":7},{"id":"7cee65f4dcd21f98ab71b0452f20426e75ca5aee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-18 15:40:05.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3349/.","accounts_in_message":[],"_revision_number":7},{"id":"b4bac44dd0b408fcdbfac659a0e222a099aaec50","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-18 18:02:16.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1164/index.html : SUCCESS in 100m","accounts_in_message":[],"_revision_number":7},{"id":"cc5d286bab0982ac0cb018dac427ec9a99430fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-25 12:59:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"73e6dfc5c39a514a2b12d27c706b54d5ab499e6e","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-25 13:00:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a619085f9a0d4644b2e618a3749695a2 : NODE_FAILURE Node request 200-0000675453 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"639770795dbd7a44155d07e7ffc0cb0016d998cd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-25 13:05:42.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62c7e661f27a4ac6a02d48614912eb96 : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0f2423a3623a47a1ae2b7a3b973ad50c : RETRY_LIMIT in 1m 28s","accounts_in_message":[],"_revision_number":8},{"id":"34d17ef7a9440c91b949e425f128efa2e92b3894","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-25 13:16:46.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3566/.","accounts_in_message":[],"_revision_number":8},{"id":"76a84ac8041cd425450a975f7ecf7ac78854326f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-25 13:17:27.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3393/.","accounts_in_message":[],"_revision_number":8},{"id":"9963dd2ff656799f1d2c96bc5c580c5ce0651d7d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-25 13:36:47.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":8},{"id":"db8665f5762ec901fa585a5e28f64a39afb9495f","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-25 13:40:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b9ca4244907dca7a0d7af4d03df25e1841873980","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-25 13:41:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/2ce88211917749aea6043692f8a279ec : FAILURE in 32m 59s","accounts_in_message":[],"_revision_number":8},{"id":"26e7c9e4b0b795225d60a123bd860fb0fd01cf92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:43:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2a30c6b1a24fcdaa5be19a53262a85\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f78015aff4cb4111b74df10d4508654b : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/254283ab63304c45ad1ce38cf2f0bc79 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e68f88e48ab4325ab06442021a776a4 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"52bebb46e1e1feadfe607191b83bc0c0bea2758a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-25 13:56:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 38m 48s","accounts_in_message":[],"_revision_number":8},{"id":"b6da5f74b9471bf900bfd2ecd01e3a40971b4cf6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-25 13:57:45.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4a64297677544c9cb517ae0debc0e231 : FAILURE in 38m 32s","accounts_in_message":[],"_revision_number":8},{"id":"9dac7e8b0bf49fc1d00ca404acdb9dae754e5374","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-25 14:13:12.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/8/upstream-check/cinder-tempest-plugin-ontap-fcp/74a5bce/ : RETRY_LIMIT in 6m 59s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/b0a8bc9/ : SUCCESS in 29m 57s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4a85ccb/ : SUCCESS in 29m 29s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"90e6ce7b2e2418f3973337ca82c5fee317df9965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 15:29:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac3cd15377e64bf1b80a39a6cf8ad407\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2937cc1c6ff94118b88315adb4eaf153 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7f7b5c6a13d4ec5921d2dd7c46eb480 : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e7a5f93f0504357a771a5c550b39830 : SUCCESS in 10m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4904c2a4df07409fbf490c93a0f5d07c : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae59ff2b59842f0bf74cf790fa598d5 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ada30c98ec484437b6a161988f3197a0 : SUCCESS in 33m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f2ca5ee31794ef38b688f69161d017d : SUCCESS in 54m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d000f42176b4c1e9034350a0a970853 : SUCCESS in 1h 24m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f5eb24e622c43d7bb210a9aa45f227f : SUCCESS in 36m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed7aa79c3e3436eb8eeb62cad01e141 : SUCCESS in 14m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f5018a374b0f455e95bab54d20d1da5b : SUCCESS in 18m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23579311c7704d65b4384a6e80589d78 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b855b8706fa4ecaa6083169042b10b7 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18c9142828824d1e9b893dea3b260ea7 : SUCCESS in 5m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1b94953991548199269ab4a1fd5e1b7 : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb6d277ef93241569fb09b1980c96a83 : FAILURE in 57m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81f895b2ba194b278697680703358acb : SUCCESS in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f030547ec9e4a658cb01534948872e6 : SUCCESS in 1h 50m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/715d41f0c0a74084b5c9e1482c226594 : SUCCESS in 2h 24m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/05a9f1556ee74bc9871b79e8cdac3413 : SUCCESS in 2h 17m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aa567da159584029a0ac8dbe711f4992 : SUCCESS in 1h 57m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24a37ea2de4d42bab5d20e8f1cd19b15 : SUCCESS in 25m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83117e3565e947dd864a743c88cfcc95 : SUCCESS in 1h 37m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d35ab88c943a424aa699f5ac76c06351 : SUCCESS in 1h 14m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/66b0b6d320674f88bc6e38b8527848a9 : SUCCESS in 46m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81bbc27cb8b2446693b254094a51c9af : SUCCESS in 2h 12m 13s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/e55931bd1c994c048f703cf1cd345426 : FAILURE in 2h 07m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7357effbdb7476b90a9fa9e52ac05b9 : SUCCESS in 1h 50m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96275bb8b4d544c68a497af12557c4d1 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":8},{"id":"67a1fae83c71e8b681f92febeb28183ebc6db43b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-03-25 15:38:28.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_03_2026_15_32_28\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"a572a9da75d1639c5236d92735980eb4682115b9","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-25 17:13:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-975880-8-fc : SUCCESS in 50m 40s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-80-975880-8-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"643bde251592a676573b926211c83a84d650deab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-25 18:07:05.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-8/FC/ : UNSTABLE in 3m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-8/iSCSI/ : FAILURE in 5h 7m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"0c3a96fa7e1909bcea0796b0fbf9b625a5c55cfc","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-25 19:49:06.000000000","message":"Patch Set 8:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"9f130cffd7121bc144b7d63b0636a9199ff3d811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-26 05:42:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"3b18b9255c3fa0d403e5ed3692109a5bcc478a3f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 05:43:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7de7ae142af94756b5f4663d58ea6e07 : NODE_FAILURE Node request 200-0000677557 failed in 0s","accounts_in_message":[],"_revision_number":9},{"id":"c4a8d42aeea5ac491cde198f9a0bb2d32b00271d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 05:48:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fe981ac1e65a47d2b35b2b04ce68e134 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9af6a754a41d4950ba467538d1a44f6b : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":9},{"id":"347b1ef279327b08a15a8b1025d020d312bcb58c","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-26 05:50:18.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6b26d9957152d6088de620359bea2fd0d8567f67","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-26 05:53:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f195785b28f9787dbd7f5c9e4fa5cf7c4699e6ef","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-26 06:06:01.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/7a3af141a3f048cf937d5fef37a41f61 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":9},{"id":"f4522682b53a0b3a52a02c91e1200345099aa15a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 06:21:44.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":9},{"id":"87e972f82c6f84d7959a87e32713f681d8d48ff2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-03-26 06:26:48.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/9/upstream-check/cinder-tempest-plugin-ontap-fcp/79fea6d/ : RETRY_LIMIT in 7m 50s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/619b42d/ : SUCCESS in 29m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/63617de/ : SUCCESS in 29m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"3729b6c0304f43489381b6e903c9b9a49fc9e0fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-26 06:42:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 38m 29s","accounts_in_message":[],"_revision_number":9},{"id":"a827d99106f82ceb6b550c742a4df6e14293079f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-26 06:44:50.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0dca0eb1d3434c939e75b31235c1d96b : FAILURE in 40m 03s","accounts_in_message":[],"_revision_number":9},{"id":"30f59c010733d16bb1de73fdf42a0e897f3197f4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-26 06:54:20.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3581/.","accounts_in_message":[],"_revision_number":9},{"id":"55ff5425cb060261a9f78c8fbc3523f731cb4d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 07:46:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd47e001e9f04aa3bacf3cb8610ae03e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa1fbbc78764da29a14ca2f45a3762f : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590a3f2c1629462ea2c01c6830a08917 : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95d4e7a509634bc6a46b17bc6eabb471 : SUCCESS in 10m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50d3a60e94144f3a91557a1b10ca47db : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3ed73440d44ac3939f969a9e3cf22a : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/0fb3c66a8d334419874cfbbe2b867129 : SUCCESS in 46m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1afdf3c767f944619a245b9c33c76a16 : SUCCESS in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/493cc4ba93614022b0c1dcafe674f5ab : SUCCESS in 1h 33m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2625614d36df4abdab925f444627be19 : SUCCESS in 1h 05m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e116f1a9ad4b4683aeeef101d6b115d5 : SUCCESS in 15m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ff971e975d04f2c8ae4149a403bd397 : SUCCESS in 20m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7dc939d1b3348e59ce1047946cd4ef1 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5602ea381a014694872f51f1f9fa0d15 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38bf962546cc45f0a3acf0fa58f5db4b : SUCCESS in 7m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cabd9bf72a474fe39e992bdeb8d4f5de : SUCCESS in 8m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f7d0dfc065b4c6a9494e199e41377a7 : FAILURE in 38m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1fdd9837dbd744548817123f041fc270 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bae906653cbe4da3a7136fb49bde77fa : SUCCESS in 1h 27m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9d1e2b3b43f48ba86b8b40da48fbc8b : FAILURE in 1h 57m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c2f84c63b3d48f18b08e94574c5bb6a : SUCCESS in 1h 50m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/67e34f8e272d48e1b5b7ef69469b9977 : SUCCESS in 1h 56m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/518068e5c97e4720b4cda7f5372a27c3 : SUCCESS in 27m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fe869d8ca5494fd8b4abce70b4081206 : SUCCESS in 1h 11m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c6301b410d714b2a8d1cda7ae78e13c8 : SUCCESS in 1h 20m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/72f33aed93214064bf85fef9e3dab4ca : SUCCESS in 27m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/75e48b33fc9346c2a1fa12a29f79c990 : SUCCESS in 59m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/18495340967040338e25a71e3c126c11 : SUCCESS in 1h 12m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c6eaeb525024484095ec151ca979b762 : SUCCESS in 1h 42m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5dfc54cf0b140fabb7113fd223346d5 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":9},{"id":"0b2c21536de5e45c4b1695f24497cd6839bf5e5f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-03-26 07:56:11.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_03_2026_07_49_54\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"eae05b0848fae27b86b58847a1161b8e160c3193","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-26 08:41:15.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-9/FC/ : UNSTABLE in 2m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-80-975880-9/iSCSI/ : FAILURE in 2h 58m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"9b22e7b121cf26ce1b1845a090472a51c53cf3e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 08:50:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafdcbf5bdc04754a3aa379889a38791\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d60aaf82334b422ea87e4b717b76cc61 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03e2030e820642f8a529725617496310 : SUCCESS in 15m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb60a21c774e497d93a8177234a3740b : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3d7ae995a18284589e4f273528b5bcaf3bb8246d","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-26 14:35:10.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1228/index.html : SUCCESS in 106m","accounts_in_message":[],"_revision_number":9},{"id":"ef429e541f4574ba17d0a31d006d47d8d8a33949","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2026-03-26 17:03:35.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"941ebb53e97eb9e7dffb8d8becd6ed216c0244f8","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-27 05:45:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"75922f5e1f98b699b49a7106cf24c176e6a5b749","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-03-28 16:29:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cba57c33ac6d8c9d3a52e81398d819374b7c9652","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 00:01:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3408/.","accounts_in_message":[],"_revision_number":9},{"id":"30424028ed6176784dc850d74bd268ea9b470d63","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-04-06 09:49:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"90ad9171d373626c5ec9e4ce5a6e3cec49492713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-04-15 07:45:39.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3f1e496869062d7718a5d4899c49316b2cec8070","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-04-15 07:46:53.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"18670edd66aaaa75e667ac95c9117a61db720c8a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 07:47:26.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6cf40ef3b6344d5a8813c55974d97c47 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/db32e0df57c546e3af68247a26f2cb54 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"f0dfed6a5f12fc795e9db9ec1e08bc96183b6e9f","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-15 08:24:09.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-fc-driver build/9b0a55d92a354c9b98307d565736bde3 : SUCCESS in 37m 58s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/975880/etc/tempest_log.log","accounts_in_message":[],"_revision_number":10},{"id":"e051bf5cea662a637c2a4f944c926bc302ce2cc1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-15 08:27:06.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/975880 : FAILURE in 39m 57s","accounts_in_message":[],"_revision_number":10},{"id":"a436799da1dbffc1b103d7746dd791252d1c1f7a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-15 08:27:38.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4b402d8ed270a1e1e1c702842c2c2004b8700392 .","accounts_in_message":[],"_revision_number":10},{"id":"1a367db2eeb085e2acca4ced8ce3323266626189","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-15 08:30:20.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/cd81112546394e2da6df0665be17f459 : SUCCESS in 44m 02s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/975880/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/975880/etc/tempest_log.log","accounts_in_message":[],"_revision_number":10},{"id":"6f6796d851a487ba257dd2aad2249bb8a958bcbf","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-15 08:32:48.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/0f9a59bd8e5e47259c6be675bfcd0da4 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":10},{"id":"0242fb4dcd46d3e54231667b2f7ae8adb3ab3194","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-04-15 08:34:05.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/10/upstream-check/cinder-tempest-plugin-ontap-fcp/2ce4434/ : RETRY_LIMIT in 8m 52s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/c07faac/ : SUCCESS in 29m 55s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/80/975880/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/99b002b/ : SUCCESS in 29m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"415f3b3de0ce389da52a9e81052df03bef0a6dfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 09:10:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a54729fa43e745de8125caeef937b1be\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/804cbaa44d2349a7ad0e7107366532bb : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1297c10af069474e9e09555040b60bf0 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac04785f1ed946b0bf0479004c94379a : SUCCESS in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dff3051fc4b569313d118bf1ca9f71859eaa5bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 10:59:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce7f0ef54fa94e5c95d4f73a589558c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6cad3acbf44c889ca4374f0a7fe6af : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54d1837bc9d648e58b733ad73f12a4fb : SUCCESS in 10m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ba634b9b340401c9a25c76a15af1c3d : SUCCESS in 12m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9dcc965fb50243feac66493708a5d4f6 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cee2040f6e4848ae689a71b9c9d1d5 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/e79f638114e14e0ead8fd9fe8ad65421 : SUCCESS in 47m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35163c856680409d83fb4e0aff0d1da3 : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/baae588a6a6046c9b157c5538052e3bf : SUCCESS in 1h 54m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a5b1e7dc42d4913a04b755fcda9398f : SUCCESS in 1h 09m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/398a7137dbda452d8fff435b612c67ca : SUCCESS in 20m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9b6a4ffc9ea43bda0973c3dbae603b3 : TIMED_OUT in 40m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cc43424abc40419eb5fcb0e42a94c9d2 : SUCCESS in 3m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/900d5d56d42c4d2995d393f3616810f3 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30dda579848144caaa38bc7c69bff5f1 : SUCCESS in 7m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f196f15001a8496d875c24bd29c215bb : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c3e5fa34baee4e75a81813cb45bc4e55 : FAILURE in 31m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0b0dfe126e60467aad9064bc2b2a71ef : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/42bd854adb4145a2a374d5ae895c473f : SUCCESS in 1h 39m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5139a5d8c9664680bdf040871dc1a041 : FAILURE in 3h 04m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f3fc50208b64b0d8e751507a1a9a050 : SUCCESS in 2h 03m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df44b085133c4f998548574731a8b839 : SUCCESS in 2h 53m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/09bb46d66dec42489f126425ce94c4f5 : SUCCESS in 23m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc78f2566a9344ef9d0d80d57c21af60 : SUCCESS in 1h 16m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fe6ff9ec3cba4c13a1416774dadf7278 : SUCCESS in 1h 32m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c833d3390d114993a2086326dabf0611 : SUCCESS in 31m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ac1edc452fbf41788853f803f6b38f15 : TIMED_OUT in 3h 08m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/221b567476ee439984383b652b0204da : SUCCESS in 1h 43m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b35639134d194b1ea61bff3cf5107874 : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":10},{"id":"d78544c4956aadb55d1d2797bd85b0cd960ec9be","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-04-15 11:07:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d95051be161731ce43e39e12434ad02a1c7cdf14","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 11:08:53.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7f7fc67324064106a201e1e7e8374174 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/618f63bb0f2146a2b460b846844425c8 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"f09d96fcebf64d0b63ce7c67bf9dd496b0bfef1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 11:34:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fdebafb03242958764da9a74cf2348\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/136b7362fc5b4c48aa6fbdf05c06b7f3 : SUCCESS in 16m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3237dec66a14b7a800186c1ed45186b : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/108ee6103cea4249801d043ee9b22e92 : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0774991a3123c9006a843452ad91464706e2368a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 12:31:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3641/.","accounts_in_message":[],"_revision_number":10},{"id":"1b0bb7f62b2ed8e86262a488ce80c1c426191f2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 20:14:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db503187925f48b6bbeb2f7b652b73c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d18fd8d60cb47f88bb492beaa56a6e1 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33f5ff7f8f0f4dadbdf8b4e19e718159 : SUCCESS in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e840ff16b9844560a2abc5c58ca47351 : SUCCESS in 9m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d63b792b73124be688f91c10d85eaf30 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfaef82c05e484ba09abedbc3071ca9 : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a0daccbb93de4bd993da432a86f9b78a : SUCCESS in 49m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e112112fe0f43c3b1922901b40a5e58 : SUCCESS in 36m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6fbb3cf145a4ae7aad2ea27ba0e5aa4 : SUCCESS in 1h 48m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba67a9e2a42c4c1387f737112372bb36 : SUCCESS in 1h 08m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab75dfb03fbd40b18f02c95dce0290e8 : SUCCESS in 14m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b1921e6bb8964de4a56c4843790ac93b : SUCCESS in 22m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f65d4295e334ce5a56af0e240e8f261 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ef4917ecb17b4624840edbbae07c01ed : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5978ff9c7a674081b0f9075b683ac4b8 : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/26423625f3604405a78eed6b019f0cbb : SUCCESS in 7m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f288317fea76429daa2555c925d1f586 : FAILURE in 43m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e62374b1db4d4c3eb4bc7f747650e78e : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e7fdacc45ec4650878778b6de8372cc : SUCCESS in 1h 45m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fdde54da2f644497986ffc0a917d5008 : FAILURE in 1h 30m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/81c131d5a6854b17bc047b42c66f28a4 : SUCCESS in 2h 40m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/efec605691d444a7ad7d8fefd01a9060 : SUCCESS in 2h 18m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0978ba6be98c464494cdb934a8342cbe : SUCCESS in 37m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/19d30e2ae52444358cf053a2cf2127af : SUCCESS in 1h 31m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/047d357f287648d180b46cce4549aab6 : SUCCESS in 2h 05m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/02db17f05e9240419ae198a62890fb2b : SUCCESS in 48m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1c91b21213c347af84879e6d6dfd9bea : SUCCESS in 1h 56m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca75f7b0acea48888a424ba77b0ccdab : SUCCESS in 1h 43m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d08d599b06a5448b8343b2c4ab1899ff : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":10},{"id":"91e8aace33ea585adade471dd95fdcbaea61196f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-15 20:23:27.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_04_2026_20_17_17\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"5bb3e7215e8505c550c1fe0e0447e83d0689f37c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 08:20:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3817/.","accounts_in_message":[],"_revision_number":10},{"id":"9e174709437d77976d9bbebc1e09eba3eab8968c","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-04-17 14:55:07.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8c20eafad90f293acf006436e5c757e602dd4ec9","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-04-21 06:08:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"03ea8dc9010238f87334a352b5dfac9022f41878","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-22 09:20:43.000000000","message":"Patch Set 9:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/975880 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"18d902a074d0b53c1c830dc65d86eb9cd4f19b67","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-04-30 07:53:35.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0e85efd77c25b578b4f5312598f8e1fd5924b7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 07:54:05.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f915cd35ee33058a67b021527311266cdb0779c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 09:33:23.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f9dc82161c4ddd8cc8192b3d7dc6ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f320050a6cd4329aeaba2fcb6a1391f : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/791b4247f12a4d65a8baa374bf4f5ea2 : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/027b3b4f1e1749ea9721d01efcbb90d9 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9167e9975bc4f32ad5d32a7405e7613 : SUCCESS in 6m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a1bd994a30a64db48a3ff96e8c3f6118 : SUCCESS in 42m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4af859f9caa34bf79a6f2299c09e5adf : SUCCESS in 44m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e76727c353d24fa0999f94dc7a1afe4a : SUCCESS in 1h 18m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/042f63b5aea743b58180c90882e791dc : SUCCESS in 1h 05m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9683f644f4f14e0c959f7faf688eedc0 : SUCCESS in 20m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b2004e35f11b4307b6fcb9e884ee5399 : SUCCESS in 37m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6634d520d3a247f99f9d649d9f038912 : SUCCESS in 1h 38m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3204399e9a784bdeaeced1efd8860a28 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":10},{"id":"6b1fda33d5799412cb560888d7712952c0d7a6b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 09:33:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"54764f1f2d556923e6d007dd69ca1c79576466ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 09:34:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef4f0d4dc94b46b198935321691de937\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1873ffe09040a294495c6dc9e12499 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a6430c7f9e4ea6a682740ba780dcde : SUCCESS in 57s","accounts_in_message":[],"_revision_number":10},{"id":"8791652b0e5560844f9bd09523470cdbd6f5b6c6","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-05-07 06:10:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2f7c6c22ee550525d561fbd4082291306766d251","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:19.000000000","message":"Patch Set 10:\n\nBuild Failed","accounts_in_message":[],"_revision_number":10},{"id":"61647ee5fcdca5d103897cdd8ef7eaa619805dbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:22.000000000","message":"Patch Set 10:\n\nBuild Failed","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"03c97a7981d512f2580577014fee8b8f5660f5b7","revisions":{"7123ceea2200798cfe926fdde27a6d39de078275":{"kind":"REWORK","_number":1,"created":"2026-02-06 05:14:01.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/1"}}},"commit":{"parents":[{"commit":"455d8710980a7ddc399f4eb579af55f6fd929f2d","subject":"Merge \"Add region_name support for Barbican key migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/455d8710980a7ddc399f4eb579af55f6fd929f2d"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:13:59.000000000","tz":-300},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7123ceea2200798cfe926fdde27a6d39de078275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7123ceea2200798cfe926fdde27a6d39de078275"}]},"branch":"refs/heads/master"},"fd7ec59c1f621b1fea7002c6a139a037e5e6f916":{"kind":"REWORK","_number":2,"created":"2026-02-06 05:23:14.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/2"}}},"commit":{"parents":[{"commit":"455d8710980a7ddc399f4eb579af55f6fd929f2d","subject":"Merge \"Add region_name support for Barbican key migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/455d8710980a7ddc399f4eb579af55f6fd929f2d"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:23:03.000000000","tz":-300},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7ec59c1f621b1fea7002c6a139a037e5e6f916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7ec59c1f621b1fea7002c6a139a037e5e6f916"}]},"branch":"refs/heads/master"},"9fef01686519d3bdab016db2943e3a73fbeabf9f":{"kind":"REWORK","_number":3,"created":"2026-02-06 05:24:57.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/3"}}},"commit":{"parents":[{"commit":"455d8710980a7ddc399f4eb579af55f6fd929f2d","subject":"Merge \"Add region_name support for Barbican key migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/455d8710980a7ddc399f4eb579af55f6fd929f2d"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:24:48.000000000","tz":-300},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fef01686519d3bdab016db2943e3a73fbeabf9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fef01686519d3bdab016db2943e3a73fbeabf9f"}]},"branch":"refs/heads/master"},"18746ae2bd756f4c451350a9c89daadc34db81aa":{"kind":"REWORK","_number":4,"created":"2026-03-03 17:50:36.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/4"}}},"commit":{"parents":[{"commit":"adbab45b06ec0edfe549f6f7f518c901b6dbb95d","subject":"Merge \"Hitachi: Allow for ADR with GAD support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adbab45b06ec0edfe549f6f7f518c901b6dbb95d"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-03-03 17:50:33.000000000","tz":-300},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18746ae2bd756f4c451350a9c89daadc34db81aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18746ae2bd756f4c451350a9c89daadc34db81aa"}]},"branch":"refs/heads/master"},"a455542ceb88a801e40880cf0d98a116fb8b4736":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-10 04:34:53.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/5"}}},"commit":{"parents":[{"commit":"d5f967dd35aea2ee7040b1ce2666a9c9374028b1","subject":"Merge \"NetApp ONTAP MirrorAllSnapshots snapmirror replication failing issue fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5f967dd35aea2ee7040b1ce2666a9c9374028b1"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-03-10 04:34:53.000000000","tz":0},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a455542ceb88a801e40880cf0d98a116fb8b4736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a455542ceb88a801e40880cf0d98a116fb8b4736"}]},"branch":"refs/heads/master","description":"Rebase"},"40a3bfc1ae579212974a5ba8982f89e8b0ca46b9":{"kind":"REWORK","_number":6,"created":"2026-03-10 06:54:35.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/6"}}},"commit":{"parents":[{"commit":"adbab45b06ec0edfe549f6f7f518c901b6dbb95d","subject":"Merge \"Hitachi: Allow for ADR with GAD support.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/adbab45b06ec0edfe549f6f7f518c901b6dbb95d"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-03-10 05:07:09.000000000","tz":-240},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40a3bfc1ae579212974a5ba8982f89e8b0ca46b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40a3bfc1ae579212974a5ba8982f89e8b0ca46b9"}]},"branch":"refs/heads/master"},"4e096c2759ecd6dd27dd38fb448382afa7a7abbc":{"kind":"REWORK","_number":7,"created":"2026-03-18 10:51:07.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/7"}}},"commit":{"parents":[{"commit":"ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9","subject":"Merge \"[Hitachi] Update docs for VSP One Block High End.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-03-18 10:50:54.000000000","tz":-240},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e096c2759ecd6dd27dd38fb448382afa7a7abbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e096c2759ecd6dd27dd38fb448382afa7a7abbc"}]},"branch":"refs/heads/master"},"5f1c62d3b55ebbe4ab2dbd165ef0038f01d99b88":{"kind":"REWORK","_number":8,"created":"2026-03-25 12:59:01.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/8"}}},"commit":{"parents":[{"commit":"ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9","subject":"Merge \"[Hitachi] Update docs for VSP One Block High End.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-03-25 12:58:45.000000000","tz":-240},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f1c62d3b55ebbe4ab2dbd165ef0038f01d99b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f1c62d3b55ebbe4ab2dbd165ef0038f01d99b88"}]},"branch":"refs/heads/master"},"90c34ca7cb427b504f87774bb517f0418d1acc41":{"kind":"REWORK","_number":9,"created":"2026-03-26 05:42:44.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/9"}}},"commit":{"parents":[{"commit":"ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9","subject":"Merge \"[Hitachi] Update docs for VSP One Block High End.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ddbffaa0ba328a5ebf53935a5b9c3659b499e4b9"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-03-26 05:42:31.000000000","tz":-240},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90c34ca7cb427b504f87774bb517f0418d1acc41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/90c34ca7cb427b504f87774bb517f0418d1acc41"}]},"branch":"refs/heads/master"},"03c97a7981d512f2580577014fee8b8f5660f5b7":{"kind":"REWORK","_number":10,"created":"2026-04-15 07:45:39.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/80/975880/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/975880/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/975880/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/975880/10"}}},"commit":{"parents":[{"commit":"a091e502dfbafeed6b791a80cc47d7da52c84d39","subject":"Merge \"Move nfs-fips job to the experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a091e502dfbafeed6b791a80cc47d7da52c84d39"}]}],"author":{"name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","date":"2026-02-06 05:09:09.000000000","tz":-300},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-04-15 07:45:38.000000000","tz":-240},"subject":"Add support for AutomatedFailOver policy with consistent replication","message":"Add support for AutomatedFailOver policy with consistent replication\n\nIntroduces a new flag `netapp_consistent_replication`. When set to `True`, all\nFlexVol volumes in a storage backend stanza are protected together using a\nsingle Consistency Group (CG) with the AutomatedFailOver policy. This enables\napplication-consistent replication for the entire stanza.\n\nFeature highlights:\n- Single Consistency Group replication for all volumes in a stanza.\n- The new strategy (single CG per stanza) and the old strategy (one SnapMirror\n  per volume pool) are mutually exclusive.\n    - Migration requires cleanup of artifacts from the old strategy before\n      enabling the new one. Prechecks are added to ensure no stale FlexVol\n      volumes or SnapMirrors exist (brownfield scenario).\n- AutomatedFailOver policy is supported only for iSCSI and FC protocols.\n- Destination CG and FlexVol volumes are automatically created during\n  replication setup.\n- New FlexVol volumes added to the backend stanza are automatically included in\n  the existing protected CG.\n- iGroups, when created, are replicated to the destination site automatically\n  when an OpenStack volume is created and attached to a VM.\n\nMigration note:\n- Customers must remove legacy replication artifacts before enabling\n  `netapp_consistent_replication` to avoid conflicts.\n\nFollowing update is done to same PR,\n1. Tested AutomatedFailOverDuplex replication\n2. Reverted back igroup replication as there are ONTAP issues with it. Will be using the old\napproach of creating igroup on both sites.\n3. Introduced host-proximity feature to enabled multipathing to AFD policy.\n\nChange-Id: I4b402d8ed270a1e1e1c702842c2c2004b8700392\nSigned-off-by: Sanketh Shetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03c97a7981d512f2580577014fee8b8f5660f5b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/03c97a7981d512f2580577014fee8b8f5660f5b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
