)]}'
{"id":"openstack%2Fcinder~976380","triplet_id":"openstack%2Fcinder~master~I2d730817ed9bc898d4808e2c80a435a32dc6facd","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-02-13 05:12:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2d730817ed9bc898d4808e2c80a435a32dc6facd","subject":"DNM - do not merge - test patch","status":"ABANDONED","created":"2026-02-11 12:41:51.000000000","updated":"2026-02-13 05:12:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe90e9a055f1b28b4927d4ac34f1a2b9c786a531","_number":976380,"virtual_id_number":976380,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-11 15:40:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 13:40:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-11 13:46:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-11 13:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-11 14:03:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-11 15:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 19:25:47.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-02-12 05:08:48.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"58314c41ff406b80c6f422f2edbb7098f2960102","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-11 12:41:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"029fd6841ba904b20da19ec4eb3d2ffe84ee8cf0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-11 13:40:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2d730817ed9bc898d4808e2c80a435a32dc6facd .","accounts_in_message":[],"_revision_number":1},{"id":"57d8a0788d0064159f8ef6445e63368750452f79","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-11 13:46:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/976380 : FAILURE in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"48aff092ecf1329de60293e74a03788be417a00d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 13:49:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39231ee894834347a0c1ceb379f51991\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f273dcdb68047e481ae65d2a2ecade7 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d079a3b5e7f45cc8f57206bff5980ee : FAILURE in 7m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9acf91a7c5754e7e896e6619064514d1 : RETRY_LIMIT in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0632da523efe837e8ab38c41cc82211f9355e3a0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-11 14:03:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/cd4b489c0b2541449e981475b86bcda1 : FAILURE in 40m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fe7ac5b9960e4dc677241b2dd5e05afa4455def9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 15:40:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bf9d3b4f3844851bf089f69b6dd73ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc2e2bf17684cd299b22a8eebb341bb : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/680f1fa8bf824f4ab73e89af2f39d1d8 : SUCCESS in 10m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e235297a74f6437a990532679ed583f6 : FAILURE in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004c02466e614c3986316a28f8e3c17f : FAILURE in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/86ed7eee58764598b41ba9f797c55621 : SUCCESS in 52m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1de27a2da0d34c6b932bfd3858e36b69 : SUCCESS in 52m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffa517d4ba93440eb42a2b24c62e2b45 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53a26ac47c3744bc9dd60fe4b5bd06cb : SUCCESS in 1h 02m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/455825e55f914a2e8d5801ab4cacf6c7 : FAILURE in 4m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b754edaebce48d1b16db64ad8e588a8 : SUCCESS in 2m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/998485d85b6e4caeb5f0f4b43bf4b65f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9793024634b844e29eb7312e694862f2 : SUCCESS in 4m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/78403e0ddd944c77b7a3b2e1708f3129 : SUCCESS in 7m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aaff44f2fa3e4f7396f45455103f6de7 : FAILURE in 28m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59fddc9ae57b41edbdc5e666788ad0a8 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be2fa8eb453745ae853f6e5d5cd67922 : SUCCESS in 1h 50m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/da7dc546de4543c5aabc133e58336490 : FAILURE in 2h 52m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f9421ed67a14b06a5eb61788991fd14 : SUCCESS in 2h 08m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cbd35412c22241edb4770b77c3203fc1 : SUCCESS in 1h 40m 44s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7e70d2fe04f1468a8e9d33cecb1246ea : SUCCESS in 22m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f1dfd0a02e6e46b18ec56a57cf5246ed : SUCCESS in 40m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0e92fd0fd6254322925ef373052e2cf8 : SUCCESS in 1h 38m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/65de3954690e4dc4a59dbebad784e016 : SUCCESS in 46m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/59d92e7e31a34737bc61e15de9fd07f8 : SUCCESS in 2h 09m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7326726c2c1a4804aa716c9f2206e026 : SUCCESS in 1h 57m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5ae6956e9d044a194ef815ff457b634 : SUCCESS in 57m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66ac5c3650974b9798cb90653db7c679 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"38966052ae0e86e954cb652f8c03dee7f5e7316b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-02-11 19:25:47.000000000","message":"Patch Set 1:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/609/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"63c3efe85eb9ee00f8ad9e944b0768dce4a8d391","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-12 05:08:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c779200d96974ddf86afcac06a8f3701 : FAILURE in 1h 00m 57s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d4fd4b0fb48b40a2836bbd355d78fe5f : FAILURE in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fe90e9a055f1b28b4927d4ac34f1a2b9c786a531","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-13 05:12:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0bc966ae0537482637ee83636654aacb5a3c8ad","revisions":{"d0bc966ae0537482637ee83636654aacb5a3c8ad":{"kind":"REWORK","_number":1,"created":"2026-02-11 12:41:51.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/80/976380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/80/976380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/976380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/976380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/80/976380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/80/976380/1"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-11 12:43:29.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-11 12:44:41.000000000","tz":0},"subject":"DNM - do not merge - test patch","message":"DNM - do not merge - test patch\n\nChange-Id: I2d730817ed9bc898d4808e2c80a435a32dc6facd\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0bc966ae0537482637ee83636654aacb5a3c8ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0bc966ae0537482637ee83636654aacb5a3c8ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
