)]}'
{"id":"openstack%2Fcinder~976634","triplet_id":"openstack%2Fcinder~master~If3328f603aff088f82546be325df08b1f3f88de5","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2026-02-17 06:19:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If3328f603aff088f82546be325df08b1f3f88de5","subject":"DNM - do not merge - test patch 7","status":"ABANDONED","created":"2026-02-12 14:36:11.000000000","updated":"2026-02-17 06:19:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"893e717c057e126acbbd7be7ee569f3765370ee5","_number":976634,"virtual_id_number":976634,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-12 17:51:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 15:11:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-12 15:36:41.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-12 15:45:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-12 15:51:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-12 16:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-12 17:51:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 18:16:33.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-13 05:42:25.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-13 19:49:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7a784efdac9eb1976ca020f46ee54e3d54c0a16b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-12 14:36:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"525ca936c2da73a7648124bd45acd4d2ebd4c262","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-12 15:11:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If3328f603aff088f82546be325df08b1f3f88de5 .","accounts_in_message":[],"_revision_number":1},{"id":"325a66ba4c927d9626ac275cd9bec7336133cb4b","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-12 15:36:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/ff322d7266d9410b96dc198ab8b7ab55 : FAILURE in 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ba4abc40b1395c544b5452c2c5667dde9e579fdc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-12 15:45:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/976634 : FAILURE in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fe04e37c9659bccc65bd63c74fac4625f7756900","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-12 15:51:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/d9114a6821454d918edead436f436714 : FAILURE in 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d64e159cd20eb9af66e3d3643ee090568408dcd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:31:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7c5a069d7a4cc4b1518b25380d7fe8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c873eacba324fd890d07c9899f55716 : SUCCESS in 22m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/947d57b696f04e2a9eccac95ab898143 : FAILURE in 13m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35f00eacd53a44bca639ad345daed66f : RETRY_LIMIT in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"556bfb45f7c4b8800f7daabb537dd3cbc128277c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:51:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d40c2aad7b9d4eabae9e49a9bba43ae5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3309ed921244f609d24a8099c1b7b3f : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1e2511640214c7c9cb28e3bebf46a2f : SUCCESS in 9m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f1655e19a404ef4bb03184110e55522 : FAILURE in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feefe8c75d454712b1ee46d25bf6c027 : FAILURE in 4m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fccb56eadb8d43fe8ac3efa1f69badb5 : SUCCESS in 1h 02m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa81c227f24c4fedb41c23d1f8f62be5 : SUCCESS in 46m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f59aeb169c6e46199e9b7b9bd2c124ad : SUCCESS in 1h 33m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a141400ae6764ebaaa52631b7aa8b647 : SUCCESS in 1h 06m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d125d4212da446a98187db78b6b443c1 : FAILURE in 5m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1bf86352871549caa6c4fb7e4f8b9e9c : SUCCESS in 3m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce2bb1e09a96472899649f2864cea969 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64f6e88175154bc5b4977bf829009d94 : SUCCESS in 3m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5deb24c3ae6642b78ca77aee817ec0b9 : SUCCESS in 5m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4cf6062e992246a0a04a625ffaea453e : FAILURE in 33m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1aac3d7938bb4cbdb95f357c78147ffb : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17ad4507acd34b85beb9d04a4370b3e4 : FAILURE in 19m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/934fcfd0411b4971b76518851cfee439 : FAILURE in 3h 00m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e43ca542e544a55baac9c8eff6680ed : SUCCESS in 1h 55m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1b59b91578324a5ca642ee469e1395ee : SUCCESS in 1h 56m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f2d570d861fc406da6c3363d175598a6 : SUCCESS in 19m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/91fe01b599bd4a2492ac2e6948afe128 : SUCCESS in 1h 10m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ca9b7c531654e27a4b61388084a8b90 : SUCCESS in 1h 42m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e458132bd24405cbce245ae5625f7fb : SUCCESS in 44m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0700014fea304ef286b6c9b5f98014cd : TIMED_OUT in 3h 09m 02s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/487807993b544a4d918c9b2134a05839 : SUCCESS in 1h 45m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ff63fecde2a469287cca4cabc513456 : SUCCESS in 58m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4284600913024c67af81ee88c90ba25d : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4f043a32d72bee5abaa42e782a763048ea752364","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-12 18:16:33.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_628/index.html : SUCCESS in 103m","accounts_in_message":[],"_revision_number":1},{"id":"1004c52958937fe1d4f44df76bc4186afa28f321","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-13 05:42:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/595a72ea074e41efa565df91e3bd0aca : NODE_FAILURE Node request 200-0000623415 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"5140fff6f6c4789d6b220d60609394840790a697","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-13 19:49:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1bb9387326dd49c2860c014c7a178d8b : FAILURE in 54m 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6793d561adde4d899f877c73b55b6113 : DISK_FULL in 1h 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"893e717c057e126acbbd7be7ee569f3765370ee5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2026-02-17 06:19:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7ccec4245fb5b0d4117b4261a2707a7301e17d0","revisions":{"f7ccec4245fb5b0d4117b4261a2707a7301e17d0":{"kind":"REWORK","_number":1,"created":"2026-02-12 14:36:11.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/34/976634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/34/976634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/976634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/976634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/34/976634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/34/976634/1"}}},"commit":{"parents":[{"commit":"d8525886bba1732f5d067c9182bf3fb9986b978c","subject":"Merge \"Fix: attachment access across project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d8525886bba1732f5d067c9182bf3fb9986b978c"}]}],"author":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-12 14:38:30.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra-uddhav.tilay@hpe.com","date":"2026-02-12 14:39:21.000000000","tz":0},"subject":"DNM - do not merge - test patch 7","message":"DNM - do not merge - test patch 7\n\nChange-Id: If3328f603aff088f82546be325df08b1f3f88de5\nSigned-off-by: Raghavendra Tilay \u003craghavendra-uddhav.tilay@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7ccec4245fb5b0d4117b4261a2707a7301e17d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f7ccec4245fb5b0d4117b4261a2707a7301e17d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
