)]}'
{"id":"openstack%2Fcinder~977114","triplet_id":"openstack%2Fcinder~master~I0790c339637f4979b886f71899d811ea5062887a","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-04-07 08:45:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-04-07 08:45:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35676\u003e replied on the change","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-18 18:59:55.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"removed_from_attention_set":{"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2026-03-17 14:26:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_35676\u003e by clicking the attention icon","reason_account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"}}},"hashtags":[],"change_id":"I0790c339637f4979b886f71899d811ea5062887a","subject":"chunkedBackupDriver: remove md5sum calculation","status":"NEW","created":"2026-02-17 15:04:07.000000000","updated":"2026-04-07 08:45:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fff3144a7b282e7ae60455bd598d9cdcab2aa54b","_number":977114,"virtual_id_number":977114,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-25 19:23:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-03-17 14:26:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-17 15:10:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-17 16:18:25.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-17 16:19:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-17 16:20:35.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-17 18:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-17 19:15:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-17 22:56:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-18 09:17:07.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-18 18:59:55.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-25 19:23:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-03-17 14:26:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-07 08:45:44.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-04-07 08:45:44.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"7918990c6bf065b804a2c0500891243c841a9343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-17 15:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b61775af843e9d6659f4a30fa70947695026c9a0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-17 15:10:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I0790c339637f4979b886f71899d811ea5062887a .","accounts_in_message":[],"_revision_number":1},{"id":"aaa646f8931e7a50e89bbcf40d5df6d9bb0b7b9f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-17 16:18:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/efa1bce36afa44729267ab385dcdc060 : FAILURE in 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9a2c06205ab22b1a2948ae8f3ca79c2743fb13ee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-17 16:19:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/763f6ca14a6d4617880abc5fb508d85d : FAILURE in 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"48ad2524e5624406be7b1fd43a996ef7a3cfaa4e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-17 16:20:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977114 : FAILURE in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"334c2f93fa6a1fbf9dce5a362e68d628a6c6d729","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:14:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33940de21e24971b010a825d291708a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9727cbec0f043328b7bb2c5d9f97d7b : NODE_FAILURE Node(set) request 7ad080b445b5454ca4f733fcc4b36bb1 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/06e039ed97af4ccc89663122b81a8a42 : NODE_FAILURE Node(set) request b9e714a7c6544858b7194270d5481586 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af13f44035dd47a3a6c32ba1f9b2eedc : NODE_FAILURE Node(set) request 9f9dac491aff47ac8f6ea953c8d4c516 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e21e15e69951fe931fff61896dc64f48c654a004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:15:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7627a4e64594869a3c2bb112d152765\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eae96969fa148a1801bb9408fbe07a4 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ef6b7195bd64cf899c039f656fc6ce4 : SUCCESS in 12m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a28264a67002494fa82869f77f06bf90 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fdf14021e54279bbed4e45f0ec3f4a : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/05c46aee26474e2595f499736a9717f0 : SUCCESS in 30m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc84044b4de5484db7b150e78f419fa0 : SUCCESS in 1h 10m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfef7e71d75a482ba9406b0ef475af95 : SUCCESS in 1h 33m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cce2cfe57b8c4b759a5dbc6006f9e506 : SUCCESS in 1h 09m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c057310abbcb4078b0ba0544d260eb24 : SUCCESS in 27m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/941edfcae7ae4faca860daef21b4dcb1 : SUCCESS in 3m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60776babca28419ba67db68260e867c3 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae8746d352d1442eb25c6e68696827e0 : SUCCESS in 5m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e1aa97e3a0a4b6d9228cf76cbe9c97f : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a485133ed704467d8e53a023ee0774ba : FAILURE in 35m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a815f54325445449e90d145a76dc9f5 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19b9eb024b1043328b1da5c8f6495beb : SUCCESS in 2h 04m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/03d8e4502c724fc9892b327b8df8758f : FAILURE in 2h 30m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9128a1b434874ada9a186c9541d2f9cf : SUCCESS in 1h 53m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/58cf90bab410455187769ba77cb355eb : SUCCESS in 2h 01m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c679d2528da24af8979751a4f97dc7eb : SUCCESS in 37m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8bb78f0250f6414e9fb42441a0377254 : SUCCESS in 1h 15m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9357df20f0f4a09823be435558a69bd : SUCCESS in 1h 31m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7cbfdf57aa1d40c09a8b4943ac434ed9 : SUCCESS in 29m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ff42b3a435574d8b9d804ef18dbcdfb8 : TIMED_OUT in 3h 34m 58s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b9a72997fee44a7a9a12ae4aeb16f9c0 : FAILURE in 2h 29m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87fce346fcb6450f9fb72f4daffe6acd : SUCCESS in 1h 52m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d5de05621d0467e828596418f72e04a : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"203fc75a5880809cc6f6891e5458355846dfef3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-17 22:56:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/004ec03dfcb24af3b206ebb5fbe73670 : FAILURE in 14m 50s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8ac1b2422d5c465fa61597ccdd478640 : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e2cb53534c50d4080d95938766aff18ab33a1a36","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-18 09:17:07.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_706/index.html : SUCCESS in 86m","accounts_in_message":[],"_revision_number":1},{"id":"37e58ca72046180e4ce43d072a560662d5c78b32","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-25 16:35:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5043c98fcc5cc36d03f505de50811f2ce941ef0d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-25 16:41:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/41e3cfcc73044a6ea90a93cea514070a : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c802243b7cd14b2da43a1f49fc605050 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":1},{"id":"24ba0d9541f4321a532de09afce30a99ddbd473f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 18:40:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf9eea6636a14d4c9b4c377f12a6065d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bf3aeb38e974787aed0af513c9351d6 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7aab4e89c60742b08293a9780d40189f : RETRY_LIMIT in 4m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b8a1c2cee8c40abb61703c3c78d02cf : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"136e8164e1b51013eb8750243df8fd4dbceed261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 19:23:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f742a17ee48c4e2f843997334f3f544c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d388414736468db94067996ccf1495 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba5230e1c580499c815f951210ceb77b : SUCCESS in 7m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c48a61ac1f94319a821434e19746e6c : SUCCESS in 7m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ef48baa47df4ed780ac480e3f7515fd : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba40b3d03424474ba9b5c06f2187d626 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/01fbe22fed4547ca844ebb4f61d5dc40 : SUCCESS in 30m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2d7777d594944f18659dd136fa9198e : SUCCESS in 30m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bdf4f2c25bdc4f7792c2a2cef8735bef : SUCCESS in 1h 33m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bafd04ea53dd4931bff1844508e8da87 : SUCCESS in 36m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ff3cf88006d64f5e918239ec74cead4a : FAILURE in 27m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5890009940348f7a72bef373bb793d4 : SUCCESS in 3m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a442d7ab3ed841f489f37bf0bd2d2a43 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddb4c75d065549f0a27c4d5daae98bbd : SUCCESS in 8m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ab1c23d950484e51990f0b2311f39f1b : SUCCESS in 6m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1ca5e96984cd48bb8557528545c08b4e : FAILURE in 19m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ffeec34f0e34ec3ba28b458b2d906d6 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ffba8dbc032423d8d81afbe6aab8af6 : SUCCESS in 1h 06m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c48bc34a7ed74942b073b5c7bafd4266 : FAILURE in 2h 35m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/67a4d8af69c04db0b073d101cc470b10 : SUCCESS in 1h 58m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c8ee8e015335467c8a446e611ae0d46f : FAILURE in 21m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ac4417c86bae4b5a8cb8b51af2c72045 : SUCCESS in 34m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/30722ef6843e469db39262025b3558be : SUCCESS in 1h 13m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/228afc387a5c4661a088cd401731be96 : SUCCESS in 1h 38m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2feb6b880be14525ab73d3ae68248980 : SUCCESS in 40m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/16fad2c0ee2e43239a2dd27e01c3933c : SUCCESS in 1h 46m 18s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/0176d31fae9e4de1a3fe53ffb8b51467 : FAILURE in 10m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/780eb5e56c694fc48e8eaaed844f546f : SUCCESS in 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0655ea90d704d2fafd0b6dbb6d94cae : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ec6c5b9d5b19dc2123c8fdf6d43b4a2370666aae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-25 19:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_02_2026_19_26_33\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"13a74c1654b98a79798a0a3f8388a7ad4c9711a8","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-17 14:26:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c010a457ff9678873faede075c5d02e3442a3e14","revisions":{"c010a457ff9678873faede075c5d02e3442a3e14":{"kind":"REWORK","_number":1,"created":"2026-02-17 15:04:07.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/14/977114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/14/977114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/977114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/977114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/14/977114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/14/977114/1"}}},"commit":{"parents":[{"commit":"143c5c12fb60ef7130ddff0fc9b9ec23bc452d49","subject":"Remove Windows drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/143c5c12fb60ef7130ddff0fc9b9ec23bc452d49"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-10 08:29:22.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-17 15:02:56.000000000","tz":60},"subject":"chunkedBackupDriver: remove md5sum calculation","message":"chunkedBackupDriver: remove md5sum calculation\n\nThere is no need for it and we already doing sha256 checksums\nfor every chunk(for incremental backups).\n\nCloses-Bug: #2142032\nChange-Id: I0790c339637f4979b886f71899d811ea5062887a\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c010a457ff9678873faede075c5d02e3442a3e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c010a457ff9678873faede075c5d02e3442a3e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
