)]}'
{"id":"openstack%2Fcinder~977761","triplet_id":"openstack%2Fcinder~master~I1b44a1395044fe4544050f9c34d80d1932707a20","project":"openstack/cinder","branch":"master","attention_set":{"36698":{"account":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"last_update":"2026-05-30 13:46:17.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38860":{"account":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"last_update":"2026-07-10 05:32:33.000000000","reason":"Someone else replied on the change"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-06-04 11:26:47.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1b44a1395044fe4544050f9c34d80d1932707a20","subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","status":"NEW","created":"2026-02-24 08:28:19.000000000","updated":"2026-07-17 21:36:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f5f5a0be33d274602c998c25d67db59b68d5ddc5","_number":977761,"virtual_id_number":977761,"owner":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-10 01:18:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 08:33:31.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 08:38:02.000000000","updated_by":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"reviewer":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"state":"CC"},{"updated":"2026-02-24 08:40:27.000000000","updated_by":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"reviewer":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"state":"REMOVED"},{"updated":"2026-02-24 08:41:58.000000000","updated_by":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"reviewer":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"state":"CC"},{"updated":"2026-02-24 08:41:59.000000000","updated_by":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"reviewer":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"state":"CC"},{"updated":"2026-02-24 08:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 09:10:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 09:32:13.000000000","updated_by":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"reviewer":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"state":"REVIEWER"},{"updated":"2026-02-24 09:46:08.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-24 10:01:49.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-24 10:13:21.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-24 10:20:29.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-24 10:24:22.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-24 10:30:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 11:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 22:34:00.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-05-25 12:27:53.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-05-30 14:27:07.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-30 14:52:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-09 22:56:47.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-10 00:13:32.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-10 17:17:19.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-11 07:18:53.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"62ebab6a8ade84746481d589240c2838586448bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-02-24 08:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6649501f5b24f6f3d41450a1a5c036de70f87f30","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-24 08:33:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/723daf19f1764939982c887ae28421a6 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8fe7dd04ce5140158b51df1c9aeee910 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":1},{"id":"f0e795b459c3ac0f8ddfdb453fc4e5550772ec90","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"},"date":"2026-02-24 08:40:27.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_38442\u003e.","accounts_in_message":[{"_account_id":38442,"name":"Amin Alimoradi","display_name":"Amin Alimoradi","email":"amin.alimoradi66@gmail.com","username":"Amin"}],"_revision_number":1},{"id":"4e795f8fc77928d0d13f4936e205c49f542f6d8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:48:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9fac2e991b4e3588bbe7f4e3374f4e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bd5c5cc59214893addeea660c673bf0 : FAILURE in 11m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90338072e4eb49f38a216abf70411e13 : FAILURE in 12m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b3c40adc47894a06bcfce8e8bcbd67ff : RETRY_LIMIT in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c31bc5be4301fe6ecf723a8de1d1ce68778db24","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-24 09:10:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":1},{"id":"16e8c61cbf141042233c858ee9f328e1066ad223","author":{"_account_id":36698,"name":"hamid lotfi","display_name":"Hamid Lotfi","email":"hamid.lotfi@gmail.com","username":"hamidlotfi"},"date":"2026-02-24 09:32:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d79217074791d62f857cf6c5cea31615775c20a9","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-24 09:46:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a98e7bc9809b461692b0eeec2e960c21 : FAILURE in 41m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c318e2986ac1d25f7a00e4de3c8bcf2f4c747be1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-24 10:01:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/126bbc0e6e4644329de46fda3232d96b : FAILURE in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"80feff6807498d8d29ccb8d43ec8fc8d0379a8d0","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-24 10:13:21.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/977761 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"38cac64ade5cc6a98ed77d55f34753e408021127","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-24 10:20:29.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977761 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c51d54bf6f51cc03ec1d5936f05a26b72efd65bb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-24 10:24:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977761 : FAILURE in 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"61b7f72d2cf14cc944597783557fc4511265b6d3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-24 10:30:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/52099e9a087140639bce7685198b6347 : FAILURE in 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0012135a2a5d802d90820b8e202a1b741a87b79d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 11:01:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f08c15c0a27749e4b99f98c665600b24\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81d327eb4094ee08873c0ae16216692 : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6a94dd5d8c447bfab5b14e0c6c529d4 : FAILURE in 10m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cae42394db2445e92b05ea04394db39 : FAILURE in 7m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f96921ddd874c5aaca8c93d7352c484 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8be41d0149422d98b33d5625f990d2 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/bf76e88440214e7ca2dab9018edaef31 : SUCCESS in 58m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad5e83be230b435bb7ae6568cb695122 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09339c110c7841c7a04c7e03aa30ca9d : SUCCESS in 1h 41m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/071a4f8273e147a9b8d2fee8a7a200d2 : SUCCESS in 1h 02m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4aa648eef7d44b2cad4f8b734f2ca89d : FAILURE in 15m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ed9789f750a844ab858f127ee6b409f4 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d383a068aa241bdb1bc7a1c53483fb1 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c3af65543ee479fae5d470987725622 : SUCCESS in 7m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f85062ab3c3d4d768d68183a9bbe87b4 : SUCCESS in 3m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1d4c37c25d1042a08860b5cb9eed4b9b : FAILURE in 39m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f82af7a96bbd469a8ccd8b9c7ca7877c : FAILURE in 3m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/382106e138db4b6f97c0caec93779ed2 : SUCCESS in 1h 01m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b1b4a9bc5e14d7da90fe99ea254b0d9 : FAILURE in 2h 10m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e9dd7a59b9244733ac2355d30252b9f4 : SUCCESS in 1h 52m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9cb0542386674f41ad7a21d2b5e43084 : FAILURE in 22m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b78c4f616fdb42668f198f80afe73acb : SUCCESS in 35m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d452753d03d746408406a4adab8e187d : SUCCESS in 34m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7096d21d5a4d4cbebde3c5afa41d3902 : SUCCESS in 1h 47m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b78895c2e97644f8957ab305059d062d : SUCCESS in 44m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ddb280c22f334d63a08ada9c7c17ea6a : FAILURE in 2h 24m 40s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/13d706d9f42e4a5d8756d35958330f7a : FAILURE in 19m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75798825490b44daa64fb5e56264f2d0 : SUCCESS in 1h 45m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea8019a7541a4181a71682c404520859 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":1},{"id":"02c988f96cf5f51365148331d08e62a445955792","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-24 22:34:00.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_839/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":1},{"id":"854edf26d90482a7eaecd96be6e92b65e0a63304","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-25 12:27:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35c56838f7449474fb47fc0af939c1c8d2b13058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-05-30 13:46:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7c1f84b46cd2d63bd38479fa1703cefa686f7d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-05-30 13:48:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ea32531a2f3437e2445cfbb3658097f4e443184e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 14:07:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19aaa7d246584ec2bd80fd2e9a044336\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c98f4b8a244c4deaac3e8c10a5a66fe9 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0e0861f929f4d0ab247a2837425438c : FAILURE in 5m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6c2ea602ace242118e6ed3b0ff0f07fd : FAILURE in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a8dc7bee6782832ae14347fd140955ec2f82db2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-30 14:09:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/0346490ee2ce498f98f283d773668698 : RETRY_LIMIT in 19m 09s\n- cinder-storpool-tempest-iscsi-multipath build/afd0758fb21d48eabd6eca65d0386e04 : RETRY_LIMIT in 19m 22s","accounts_in_message":[],"_revision_number":3},{"id":"27aa51dd8c3f60e9c8264263593513aff58ee060","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-30 14:20:57.000000000","message":"Patch Set 3:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977761 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"9a77e4c1cea8ad21f237a17f5924ff173e7e0b9a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-30 14:27:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver build/5cd170cd95594e66a6b17205e57fef0d : SUCCESS in 36m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977761","accounts_in_message":[],"_revision_number":3},{"id":"dfc6fa5220ab3392f085bce3d0d7167b6070fd21","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-30 14:35:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b4cb466dca324d9eacda829476f4221e : SUCCESS in 41m 30s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761","accounts_in_message":[],"_revision_number":3},{"id":"630c5c28fc19a26ec3a49cecb64960d01d784320","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-30 14:39:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/e809794f74894983a2c1d2f93fb8d7ca : SUCCESS in 48m 47s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977761","accounts_in_message":[],"_revision_number":3},{"id":"ca4644fefc5ccdb7ff0847900b7859148b4eaf4b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-30 14:43:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":2},{"id":"bf080b427299f3710c82de348787436f31f435d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-30 14:46:37.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":3},{"id":"17788ed34c69f50292dc0ddcd583e9c867d4a99d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 14:52:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4120/.","accounts_in_message":[],"_revision_number":2},{"id":"6946dd39040d54e334fcab289346f89b13bf4efa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 14:53:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4296/.","accounts_in_message":[],"_revision_number":2},{"id":"179585d35ce4ec78a73272a190ff632f01496472","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 16:16:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4297/.","accounts_in_message":[],"_revision_number":3},{"id":"bbe674e3d2a92068cee8a5a08def21277b6aaefb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 16:22:48.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4121/.","accounts_in_message":[],"_revision_number":3},{"id":"55cf8222d438fbaec834fe06ec6096f2f6ebaa5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 17:43:55.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7001c8ced8f43188ef7cb57a469863e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23525b2303ec46c1b04129ee9088df40 : FAILURE in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7b2bae4b3b540dd816daaff9b2d65d4 : FAILURE in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bacda6aaeab34eadaaa7c2e8e16a772e : FAILURE in 6m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b46f52fd751b43478390e2b72e65dde5 : FAILURE in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23678f292e95432d8e7920eec3f97b1e : SUCCESS in 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/3c568e13e5fd4e46bfceec6941b5f369 : SUCCESS in 1h 28m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1d048fcec1ca41ffbe545d8e5621b251 : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cecd0faba7fe4a4ea3988ecdd2c43e80 : SUCCESS in 1h 42m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/794b9025601b4296819e0488e3fad248 : SUCCESS in 49m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9517456b01274ec2888369f2666e0f70 : SUCCESS in 9m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5479ea3ec0a0408faced0abd9ad00786 : FAILURE in 3m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/868efb72e9124e19ab57faaed4a5f362 : SUCCESS in 4m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87a0064ef54d44deb3d850df7bf124ef : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/699c96d2ccab46a68444d2c3f77883c4 : SUCCESS in 8m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b0bc60776d542b2a9b9a2971efbe547 : SUCCESS in 6m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/78c6a254775e4ec7ab739810e4b837bb : SUCCESS in 9m 09s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f32e28c46d24fa58e9b2e0060df7141 : SUCCESS in 30m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c367ab38170421082ab445bed3ebb40 : SUCCESS in 3m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a1ae6b7dba34597843f8b5c43abfa20 : SUCCESS in 1h 56m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/988589d1e2bc413c9386b2ca5fd8a920 : TIMED_OUT Host unreachable in 3h 49m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6fe26e9486fc422398853bebc758bdaf : SUCCESS in 2h 29m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9402efec6bd345b08933c155ae74be88 : SUCCESS in 1h 25m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c7f72015587146a9917e66f7eb778461 : SUCCESS in 23m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/65fa32388cbb42adbfad20f90c478063 : SUCCESS in 49m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0d32610696d0417482d11628e6333e43 : SUCCESS in 1h 44m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88ca076e6a39475d985832dc5fdc7052 : SUCCESS in 26m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a7b5ddb01b44fd9bf6f0a203cf09133 : SUCCESS in 1h 16m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cccd0e06be7940f3b075e222df45425a : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4159b66491647ec7247f7730395b7caeb1f5cd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-05-31 07:39:19.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fcbd9b632fd6d4481b9fa3200165c82b67bba330","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-05-31 07:40:57.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0431fd75c0275a619199bbe9a4850343f31635b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 07:50:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bf705e8c97341549c64bf15b8349b5d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f89250fba124f80831d02cb8cebc423 : FAILURE in 7m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a187f368087c429b83d15196ae96cb1d : FAILURE in 7m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da0d50f926a946fe88a9e41bb7de5a5e : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af9c092c46b7a1e8ed43476a78d841b8605ee7cd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-31 08:02:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/54a3da5ee18e48d2934e36d4e06d3c35 : FAILURE in 22m 39s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977761","accounts_in_message":[],"_revision_number":4},{"id":"78e4cb402f8820dd716e301fd484344d7f92bf84","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-31 08:03:37.000000000","message":"Patch Set 4:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977761 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"0dc7293aa9f5bd6b8b12dcfdad53f5165427a2c0","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-31 08:18:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver build/eb6d5bf147e64726bbfabf1fffe4e7fd : SUCCESS in 38m 26s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977761","accounts_in_message":[],"_revision_number":4},{"id":"b6d00956252c07c250315a29c32943cb91d2f47c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-31 08:20:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/9c4dc03b71634a2d88b789f6febb0a9b : SUCCESS in 40m 23s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761","accounts_in_message":[],"_revision_number":4},{"id":"41cc685a848dd109cccf0bb9bde8266b77be6e46","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-31 08:40:30.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":4},{"id":"11f8ee13b2fd34ff43da19b00515b7c82aac7749","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-31 08:44:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8b4bb17c7cb04a1793a822ff0f8c46fb : FAILURE in 56m 36s\n- cinder-storpool-tempest-iscsi-multipath build/f9f22d9b86614857b652330d88bdb92d : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3442d061f3dbad49da63e8c71c793a853e6fee6b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-31 08:48:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4300/.","accounts_in_message":[],"_revision_number":4},{"id":"78a7af8653bc578b3aad2e0a9c483b5c5034f17c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-31 08:51:21.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4124/.","accounts_in_message":[],"_revision_number":4},{"id":"01febfa4ec5b68f70074023ba0d69e1da8b70962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 11:26:55.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaca4b96d6b64087815d428539fd32e3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af4919b62ce4739b78bbf3564e04734 : FAILURE in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca5e5ec5899a469e90cdab130d80fbd0 : FAILURE in 6m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdcaafe1df904d2ea7c8036ee0edec2b : FAILURE in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e09a5765dd8d42feac32783140a6bcd0 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b1e4a9a314450d9a79ddc90b619755 : SUCCESS in 15m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/fc3535d2f3294cde9fec08ddd1a8222b : SUCCESS in 1h 05m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dab3463e27e348658370a12e8927d5d5 : SUCCESS in 29m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8be58cab3f5c4edba9ae2aa3548919ae : SUCCESS in 1h 34m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8051d63b6018427aa234fdfc9ebbd370 : SUCCESS in 48m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a84ad2043fc04f7a891c5f65b98be35f : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9c70421ac254d0686a1a3f7d15fbc6c : FAILURE in 5m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/73fdcda8f3ba4690840a332bca2a33d8 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91f01881a3b045c5923b6c188b9ef1e4 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6d01f842ae9468293cfacfef52f69d7 : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fa07caec8a0646b2b4c1e7d0786050f6 : SUCCESS in 5m 32s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b3378b16e7c74fbeb3c19048f55ffa6c : SUCCESS in 7m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed4fb4034fc24967a4a22f815bc38ecd : SUCCESS in 43m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3bf3ea7165854aa3a0305ef518d6ed9a : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b258bb16f41048128b33896e04cde6ae : SUCCESS in 1h 55m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5220afcb8825495288aa281188869a22 : TIMED_OUT Host unreachable in 3h 42m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b06de45a888d4a878efc24387ec81822 : SUCCESS in 2h 00m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/930b565fa215421f988e1fceda15d05e : SUCCESS in 1h 29m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/afdf90bfc3764aaa9e8be3ad918804d7 : SUCCESS in 37m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49385f642c9b4672b68a01dd298d8c0a : SUCCESS in 1h 14m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/107c45032e4e44adb0b841e84ffce77e : SUCCESS in 1h 40m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7d64b0b1ff4843b78db2d0880198b408 : SUCCESS in 30m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aae69be655fb4be8b2af9ed62bca46bf : SUCCESS in 53m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7727f1ae1f24718ba4b9540ab7e9f0c : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":4},{"id":"58d00be5568e9b4b3f73daac9bdb2fab000781f1","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-02 09:44:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2b0231bc95611086ea63179f962f62f9b696fbf3","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-02 09:47:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e96e24fec87598d2bfddcbbdf5f9c01cd4439726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-06-04 11:26:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b981eb5f1fd5a00f5fa0544dbb50d027f9e4922d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-04 11:35:37.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4357/.","accounts_in_message":[],"_revision_number":5},{"id":"a1d54b8d2d89213613d45f8a63e92c598e99e8b1","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-04 12:04:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/40ed4a01e4084b368ce939c77f15c175 : SUCCESS in 37m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977761","accounts_in_message":[],"_revision_number":5},{"id":"f8f74ec7bbbb290342ae724f2b6623ad704f48e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:06:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51d8ff7a647484799ecfb04dd7053c7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b772b5a93f3248278a881d6cba81602b : FAILURE in 19m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e478c84d60a49ba81e4d9cc05285a49 : FAILURE in 13m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0ddaf028f9ab45cb91cd92cb8d46f905 : FAILURE in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d2110882c11c1bb3d3c40c5468337e98ad7a3405","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-04 12:08:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/324a634a4c124da7aee87e5a1b565663 : SUCCESS in 39m 28s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977761","accounts_in_message":[],"_revision_number":5},{"id":"0d0d51e2d3d4711dd0022cf4fc593ea4100ab731","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-04 12:14:31.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/708ed5e795c848f68611f54ad7984ffd : SUCCESS in 47m 19s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977761","accounts_in_message":[],"_revision_number":5},{"id":"8ee57bf1461a8e6ad7d9304cf9808aeda9f89a81","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-04 12:21:21.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":5},{"id":"5ed0327e0e12849fed273de810bf75a4d18813c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:40:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1800445e26a4239a1c8a5d3dcc0d002\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2fa4413eec45a09e9b21cf2def1134 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7992f0689f7549738af9a1180df06441 : FAILURE in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90a4349832094a4eaa0e99b7ad31e8f2 : FAILURE in 10m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0fe498bb9861407db378cc01fcf47861 : FAILURE in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eebe18a159f41c9ad4edce54a79a859 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/9f623453df6c45a7b8a578b7028b054d : SUCCESS in 47m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21d442a1aeef466d844791f2c4bdeac4 : SUCCESS in 48m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8a13abdbc6f4afe92f6ab93c0a3e25d : SUCCESS in 1h 48m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ba7b89506b84c06bc22907490c3332d : SUCCESS in 34m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7412bf6b31954539a94db95203637ea3 : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6004440a00424340a492c0033464521f : FAILURE in 25m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d0e3ae9ab8d4d949c1c7f5ce425aa2d : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d5ef41828b2d43da8b0dacb9819bee64 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/597a488659ca4cfebc7dca7a6a583f2a : SUCCESS in 6m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6d130f2891d94e26ad6532e2b30ad40e : SUCCESS in 6m 01s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3d90dd1050df41d994b64bd7b2cbde52 : SUCCESS in 6m 22s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/eb1ea77f95ac4dd890540e73d6c6e05d : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/97ced0ece9cd4ea2804cbfa67f119f92 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f327023bdc524f1886284351df6b9e53 : SUCCESS in 1h 37m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb95a22615bf49009ec72f7a867a57a9 : FAILURE in 1h 34m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9bda446c49248d6a16249b922934305 : SUCCESS in 2h 06m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/05296d705c08408eb5d81f3f382994aa : SUCCESS in 2h 04m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f6d8c3145b894ea4a851be4c3fc89658 : SUCCESS in 38m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c8bdf35960e4a128e4fecfc6a7d61ee : SUCCESS in 37m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3521bb1d294745f696ae8d946ea284e0 : SUCCESS in 1h 43m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eb7c7077feba4e2893710e730b9d6ce1 : SUCCESS in 42m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cda0dc3b007e4f4cab229858f115b59b : FAILURE in 1h 37m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5146f7971c54fbdb85116bf2a439a8e : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":5},{"id":"eacded6b626bf25e10e7c4b1b39cc5a11c344bf5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-04 14:18:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/23d1c4ae69bc42f48197355ab7bfadba : SUCCESS in 53m 54s\n- cinder-storpool-tempest-iscsi-multipath build/1fc17decb6764770a30397f5f9b2ea08 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":5},{"id":"c6927d856746e51911f43bfa73e277322d28b1c1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-07 08:52:59.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4181/.","accounts_in_message":[],"_revision_number":5},{"id":"19194486d3345066505ef129e27e4037b29d385f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-07-09 21:54:12.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"33e6ae432ab644e310dc95e1015b9a63ab452b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"date":"2026-07-09 22:15:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2d7f5c2bc1aa894494d7783b423bc68e43110b39","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-09 22:56:47.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ab306e30c9174f46a43c32dadd874228/","accounts_in_message":[],"_revision_number":7},{"id":"0195a1dc0a0a614903aa813be7027e812f5afa1f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-09 22:57:44.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":6},{"id":"110e459c9a4aa15bf2d7254acb514a0c58bd5cca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:05:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8178327041459dab2992b67ff9f7e5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83a7e59d4efd490996d05e4b98850a0c : FAILURE in 18m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/16f87f73bb6448b989d0f9aacb12a5e0 : FAILURE in 20m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cd46304040714bcabca0729449f1b7dc : FAILURE in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"77786c98d9a188b14f8e94e6bed3cc75e2cfd31e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-09 23:16:07.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1b44a1395044fe4544050f9c34d80d1932707a20 .","accounts_in_message":[],"_revision_number":7},{"id":"4034f727b1095e95bdbacc48f9f789d6862d211d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-09 23:27:49.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ab306e30c9174f46a43c32dadd874228\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1940ec36bced40948ee9d56941dfab1b : FAILURE in 47m 43s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/410cc1f26340411dbcbeeafa8622bace : SUCCESS in 39m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3e399f4f6844c0fa0b5e500ac9b8251 : SUCCESS in 48m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08ff465e789f48acba0af244f1875dd3 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":7},{"id":"6d6cb2ab865bb76bd0dc0f2afd3609246707b6c8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-09 23:54:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d7cbf55313334d5980b83482f9a6474c : SUCCESS in 55m 57s\n- cinder-storpool-tempest-iscsi-multipath build/95d995681c764f73815ae37cbeef815a : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":7},{"id":"5ac3002f200c539cc44af9812706ca7fd00d5686","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 00:13:32.000000000","message":"Patch Set 7:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/834 : SUCCESS in 0h 42m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977761","accounts_in_message":[],"_revision_number":7},{"id":"983911ede25dab2923a1d96e7eb0fb50972b10d4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 00:16:32.000000000","message":"Patch Set 7:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/478 : SUCCESS in 0h 43m 18s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977761","accounts_in_message":[],"_revision_number":7},{"id":"f2def2af03f18b5feda682e7ef75b9b629d9c94e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 00:17:37.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/564 : SUCCESS in 0h 44m 33s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977761","accounts_in_message":[],"_revision_number":7},{"id":"c43bd9008ab4e331d79c107387341d19adc17f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 01:18:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60d04611e51149bf8748a237c024d9ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e203b927f0403caddd654b00caa5e1 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55351fc0c4fd4237a46e9d4bf61880e7 : FAILURE in 9m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9ee662261ba4f4ca8ae8cf2522b9ba2 : FAILURE in 8m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aa278a5033d748e1a3e36fcb54b7f5ee : FAILURE in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c599b1e80047f8b78f8e1b01b36fc9 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/831d358ccfb94e218b0bdebb61d28e6c : SUCCESS in 55m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e796ae189e71456db41efd5104132e94 : SUCCESS in 37m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d3d201930874d368cb152821f3d50e4 : SUCCESS in 1h 31m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded5bb6dc1404eb9b79fd47ffc3ed451 : SUCCESS in 1h 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f2d397ffc8c422e8c30b40aef4c1f93 : SUCCESS in 9m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ee641ba4ffc04aa69ef44253065e1082 : FAILURE in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/377f9688e7124dc4bc23cc330cad0c68 : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e131e98f1ef437da3e691de3213326c : SUCCESS in 6m 36s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/73508398da1c47afb82dcd6738b6d24c : FAILURE in 13m 25s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a564dd523e6a49c393aa39cb8a1ed80a : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1c9a4007b25c47ddb4831c21e339fd42 : SUCCESS in 5m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c829bf6ba00e4459a1d2ee98eeba1841 : SUCCESS in 6m 55s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/627741e1f34a4b5faebfcc662f0b3117 : SUCCESS in 28m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e8dfb6a6d464557a2ab43ef84f1c4be : SUCCESS in 3m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7874caa3a0904b7ca0bec250b33cb3db : SUCCESS in 1h 10m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28bbd25037404ea1aa9e7388bf6f05cf : FAILURE in 2h 56m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cddcdc20a7a0471889c4af9ad00ed230 : SUCCESS in 2h 11m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bfd938e451cf446cae401d3d4ba8f7bb : FAILURE in 3h 00m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df4f9875a81f4e3f957c9b3bfbbd7073 : SUCCESS in 2h 00m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/616a0af12fd34457a6b11d107b6aee0d : SUCCESS in 37m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/420883c200204352a34f0e3af9ae961f : SUCCESS in 1h 17m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/02d7472c86ec492987c6ff70a91633d1 : FAILURE in 1h 28m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/25ca22bfaca44c848b2b228ef0402de5 : SUCCESS in 28m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a01134b678a43039e701708b31cd905 : SUCCESS in 54m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5540a20da264fc796fa2441ae1b8587 : SUCCESS in 1h 01m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"0087b9df8bfce541797b35f6cb95541e1418bab6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 05:32:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4754/.","accounts_in_message":[],"_revision_number":6},{"id":"e311dd2d108695acf2387c8d99946b8b6f9cecb8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 07:52:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4756/.","accounts_in_message":[],"_revision_number":7},{"id":"e4740b5499f1f377de5fd17b088cd04138adcd7e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-10 17:17:19.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-61-977761-7/FC/ : FAILURE in 19h 1m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-61-977761-7/iSCSI/ : UNSTABLE in 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"15f4ade85439919b3a9fb47c68729c8580e0e6ca","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-11 07:18:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-61-977761-7-fc : SUCCESS in 1h 37m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-61-977761-7-iscsi : SUCCESS in 1h 14m 59s","accounts_in_message":[],"_revision_number":7},{"id":"449ff8c0847fbd62bfa734fe47452110e8fd2e00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 21:16:31.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4578/.","accounts_in_message":[],"_revision_number":6},{"id":"f5f5a0be33d274602c998c25d67db59b68d5ddc5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 21:36:53.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4580/.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"46b8fc747a0f5b8b79b91caaeb821718dfa376a1","revisions":{"d605fd1573c7dbf9e4a4ef3c1041c51dfe47de2b":{"kind":"REWORK","_number":1,"created":"2026-02-24 08:28:19.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/1"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d605fd1573c7dbf9e4a4ef3c1041c51dfe47de2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d605fd1573c7dbf9e4a4ef3c1041c51dfe47de2b"}]},"branch":"refs/heads/master"},"d66f6f3d3365b72d1787732023f5e7b996345b66":{"kind":"REWORK","_number":2,"created":"2026-05-30 13:46:17.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/2"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheillinred","email":"soheillinred@hotmail.com","date":"2026-05-30 13:46:11.000000000","tz":0},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d66f6f3d3365b72d1787732023f5e7b996345b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d66f6f3d3365b72d1787732023f5e7b996345b66"}]},"branch":"refs/heads/master"},"402239ba9009e60c3fa07a7cd4457c83e6e00fca":{"kind":"REWORK","_number":3,"created":"2026-05-30 13:48:59.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/3"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-05-30 13:48:45.000000000","tz":0},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/402239ba9009e60c3fa07a7cd4457c83e6e00fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/402239ba9009e60c3fa07a7cd4457c83e6e00fca"}]},"branch":"refs/heads/master"},"2695739c4454feb190782f57ac10fa5c3390ccd5":{"kind":"REWORK","_number":4,"created":"2026-05-31 07:39:19.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/4"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-05-31 07:39:19.000000000","tz":0},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2695739c4454feb190782f57ac10fa5c3390ccd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2695739c4454feb190782f57ac10fa5c3390ccd5"}]},"branch":"refs/heads/master"},"11a11900658e1d13233453d86bbdb707c3c73e33":{"kind":"REWORK","_number":5,"created":"2026-06-04 11:26:47.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/5"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-06-04 11:25:57.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11a11900658e1d13233453d86bbdb707c3c73e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11a11900658e1d13233453d86bbdb707c3c73e33"}]},"branch":"refs/heads/master"},"c6e1f3e78fea5ba46b2dc5c8fcb52e057c4ac8df":{"kind":"REWORK","_number":6,"created":"2026-07-09 21:54:12.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/6"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-07-09 21:54:08.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6e1f3e78fea5ba46b2dc5c8fcb52e057c4ac8df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c6e1f3e78fea5ba46b2dc5c8fcb52e057c4ac8df"}]},"branch":"refs/heads/master"},"46b8fc747a0f5b8b79b91caaeb821718dfa376a1":{"kind":"REWORK","_number":7,"created":"2026-07-09 22:15:31.000000000","uploader":{"_account_id":38860,"name":"soheil","display_name":"soheil bagan","email":"soheillinred@hotmail.com","username":"soheillinred"},"ref":"refs/changes/61/977761/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/61/977761/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/61/977761/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/61/977761/7"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-02-24 08:27:44.000000000","tz":210},"committer":{"name":"soheil","email":"soheillinred@hotmail.com","date":"2026-07-09 22:15:12.000000000","tz":210},"subject":"Fix: avoid automatic checksum headers breaking S3-compatible backups","message":"Fix: avoid automatic checksum headers breaking S3-compatible backups\n\nS3BackupDriver currently sets request_checksum_calculation\u003d\u0027when_required\u0027\nin the botocore Config. On some S3-compatible backends, this causes\nput_object requests to include checksum headers that are not supported,\nleading to backup failures.\n\nCloses-Bug: #2142548\nChange-Id: I1b44a1395044fe4544050f9c34d80d1932707a20\nSigned-off-by: soheil \u003csoheillinred@hotmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/46b8fc747a0f5b8b79b91caaeb821718dfa376a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/46b8fc747a0f5b8b79b91caaeb821718dfa376a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
