)]}'
{"id":"openstack%2Fcinder~977767","triplet_id":"openstack%2Fcinder~master~Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0","project":"openstack/cinder","branch":"master","topic":"bug/2142083","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-19 13:09:15.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-03-19 13:09:15.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-03-19 13:09:15.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-03-19 12:39:00.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-03-16 18:12:49.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0","subject":"Dell PowerFlex: Fix san_username rotation","status":"MERGED","created":"2026-02-24 09:29:54.000000000","updated":"2026-03-19 15:52:49.000000000","submitted":"2026-03-19 13:09:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977767-bug/2142083","meta_rev_id":"a0cd697733890e6fb0914c3681ef3a07d58c0bc0","_number":977767,"virtual_id_number":977767,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-19 13:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2026-03-19 12:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-16 18:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2026-03-19 12:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 09:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 09:47:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-24 10:11:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 10:35:23.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-24 10:58:17.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-24 11:04:47.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-24 11:06:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-24 11:14:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 11:23:29.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-24 18:26:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-02-24 18:26:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-24 18:26:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-24 18:26:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-02-25 00:16:37.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-12 23:17:03.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-19 15:35:14.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-03-19 15:52:49.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"dbc9a738ff6d48bd960b633bac69e2440d5c1ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-24 09:29:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82c3415b24e10fa990032fd7e29fd7cfb935089d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:47:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7e87b1b1914e59b38097c0e0f16110\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe65700f4db341579c82ca79fb831550 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4a9812e3ff43baa39012f2c8ac2660 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a6cf9288a6540b5bc53aa8ae315d596 : SUCCESS in 9m 30s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/658b7c36ce654103bbae177ad28d746e : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"68e9591e618c21ea62d834258c15e4e13b6aa4f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-24 09:57:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_02_2026_09_50_31\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"d2b17369143000c25b1ab890f9ed0a8529262a4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-24 10:11:21.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0 .","accounts_in_message":[],"_revision_number":1},{"id":"dd25fe9fa01117f1c0ed883288d8dcc2a7c9a8e5","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-24 10:35:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/0377f33622014b338c467edbeaf558b6 : FAILURE in 32m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0708cd0b9ecf43ba76be471fc846dcb764103f9d","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-24 10:58:17.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/977767 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8d0a298d3daa096f585da128328b0c8c9f82d1b0","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-24 11:04:47.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977767 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"108727400cb1856fd08193e6fdb0740e144b0ddd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-24 11:06:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977767 : FAILURE in 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7311c0acfc28f837cc3b01060eb0d038bf288726","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-24 11:14:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2237f20eaa994b17b053f146ed460838 : FAILURE in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"98f8655495d4b248ee0c1c0dfd87da0d910cef4d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-24 11:23:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4f5e3870d6ad4499a6cd47a54763d8b3 : RETRY_LIMIT in 12s","accounts_in_message":[],"_revision_number":1},{"id":"c8ef536b16475e9f2669d89007d609cb540db2b3","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-24 18:26:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"752874cf22c131e894fe93bfa33a3a774a5d1fc6","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-25 00:16:37.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_840/index.html : SUCCESS in 102m","accounts_in_message":[],"_revision_number":1},{"id":"62ec291926de546f09eb4f89ee9afc958b3ed411","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-12 14:09:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c0ec87c3a676fe5cc1adcb33a8d7ad44204c9f2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-12 14:10:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"952d65a22849b862106584fa6a4022714de5c698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-12 15:12:53.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"408aa25eb395e3b91f77674375b8366c8ef636fe","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-12 15:16:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/2a05c65bbe78441a9de2536cac159971 : RETRY_LIMIT in 11s","accounts_in_message":[],"_revision_number":2},{"id":"4de049bd1df34b1b34ad97cfe669631dbefb6be0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-12 15:17:34.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d55154fc8523ea4b2aa39d96df520ee7df813245","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-12 15:19:20.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"de545294149cf5dcf393fc5ef97f8e4dc28aaf33","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-12 15:29:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5b0edc80c8df44d999374bb45764dae9 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"038af3727bd98b13443d737af0a40e5b1300a722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 15:37:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb3fd7e373ff4d1c9c9a53b9d79cd17d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5fe6877ed84eaba543707a3c86197b : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adfb27dee234eac83ce65ed87a2fdee : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bff25182354461885caa5f8839296a3 : SUCCESS in 12m 06s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/72add30478344d989f74b0cfb329d2ca : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2946f42637a540671582371c8a1d26642834863a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-12 15:50:06.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0 .","accounts_in_message":[],"_revision_number":2},{"id":"1e9d31c4af95de8e9c4110ba7660281731d68ec1","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-12 15:51:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/6de1f2a1847948b2abcb38908058f644 : FAILURE in 32m 22s","accounts_in_message":[],"_revision_number":3},{"id":"49181ada1ac74bcf6af1cad6e025b2ed0c6bb561","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-12 15:55:18.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0 .","accounts_in_message":[],"_revision_number":3},{"id":"4da1a00c8f3de15a0e330369a0da351805dd9b72","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-12 15:58:37.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977767 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b6e5eee0e04693b2b63cdaaf379e82ab364a5387","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-12 15:59:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977767 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cf7b593b092d76e0bf00b04c5b8ad43b55e96bc3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-12 16:01:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/9232614086524f0db4f717cbaf86432c : FAILURE in 38m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c227e9b1afc28f1fc6a0d4f19a9da56adbd6f116","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-12 23:17:03.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3464/.","accounts_in_message":[],"_revision_number":2},{"id":"2a1e8a227419dc794a591c75412dfdeb0d1b7a1b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-13 00:40:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3291/.","accounts_in_message":[],"_revision_number":2},{"id":"f9fecc31cdff09ec33cc444086ce76825b8eca63","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-13 00:47:50.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3465/.","accounts_in_message":[],"_revision_number":3},{"id":"fec00c4565881da41ec8c88d0e20a80cb04d92e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-13 01:57:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3292/.","accounts_in_message":[],"_revision_number":3},{"id":"15d7ca27600df0d178a0a009d403c927849212bd","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-15 18:52:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f9d64dfe157c052da8b9e18fe12eb7affce4a28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-16 05:12:46.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f563f826918a27926589bde4c5136c8595950e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-16 05:15:32.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"19044c79bdc6c98152593004fff313dc70d7be17","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-16 05:16:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"86cc7e043b401ea1c59ee8ead7dbc5ae1a8a0309","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-16 05:19:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4f15ad04f6159a4ed8e72a56fce7a5e3cdb3489f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-16 05:29:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/b6250f8edcf44c88b29e535ff0799c6e : FAILURE in 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4425eacba4d50f368621e1c52f5307f96647e4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 05:33:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c002479ed5442f898c2acd303072067\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fcb88d0ad24901826eec23f278fce4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb1c4d7148a43fdbb7ef1ef15217105 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2370f7def5b40ef93a80d0a87445951 : SUCCESS in 11m 21s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bdbff9cfb27d49738b50414c272f24ae : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":5},{"id":"65f7d23dc025ff6a301b58e1a88e20b59c716d58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-16 05:37:37.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0 .","accounts_in_message":[],"_revision_number":4},{"id":"4c15a1392d1bffc163b774b8c8c6bde177f11a1a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-16 05:40:12.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0 .","accounts_in_message":[],"_revision_number":5},{"id":"2a0bc4db6806c35384ca22cc61ee0b95aac56cea","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-16 05:43:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/c03ee229d63f4a2abe346c0681050c66 : FAILURE in 26m 23s","accounts_in_message":[],"_revision_number":5},{"id":"002fac019070713f4b1db808a65d96443fd50328","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-16 05:49:28.000000000","message":"Patch Set 5:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977767 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"9464f79ec90413a57d72e677dc23c428d7f771e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-16 05:53:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977767 : FAILURE in 33m 48s","accounts_in_message":[],"_revision_number":5},{"id":"04954c96137e02ac3cfdb34f9e54f06f159d4deb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-16 05:55:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/06d120283113459f8d21c4f2513cedd9 : FAILURE in 34m 36s","accounts_in_message":[],"_revision_number":5},{"id":"e8a9769f68f819148ef56f621ae8830a501d7f17","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-16 06:21:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3487/.","accounts_in_message":[],"_revision_number":4},{"id":"7cd6dcfdd4f89ce15c1a4c01fe48897d316feb69","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-16 06:22:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3314/.","accounts_in_message":[],"_revision_number":4},{"id":"db4d022a9156f556dce21341317d9b8cf1523587","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-16 06:53:33.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3315/.","accounts_in_message":[],"_revision_number":5},{"id":"f116e552f697a767f21cab528ff8344d171802a8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-16 07:31:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3488/.","accounts_in_message":[],"_revision_number":5},{"id":"afaa7dc6808779b10fee1213768373a9e85be1de","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-16 18:12:49.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac1575e63e0fc1f6986f1f8679bebc89a083a695","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-19 12:39:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"abacc453d2020aaec5370f62e770b33a2e210e1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 12:39:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0a4588912d916934fb114adb14229f848ff43f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 13:09:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64a25da42eea4525aa18f584d0414994\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965d66b01c1741d280aa1a7510595c55 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862718feee4e4c3c878c8e9d991650aa : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77f59fa568c4191aa9d0ef710bf5762 : SUCCESS in 19m 44s","accounts_in_message":[],"_revision_number":5},{"id":"1da2aed19c192b0888e8a196ae2c38a9013e4b4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 13:09:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c71e44168024ff44813f64ceff7f209df0c7e84e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 13:10:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c46b202189f47d882f8e705dd6a4bfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63eb703e1e164bf8980b9435a9712656 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae6f2802991c4b1eab2753b4a66c9863 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5},{"id":"6dec3e16e456bdf909a254ad572ca30a51b38924","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-19 15:35:11.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1179/index.html : ABORTED in 70m","accounts_in_message":[],"_revision_number":5},{"id":"f8c0f94926f76835e819345f6fa64a39a481e135","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-19 15:35:14.000000000","message":"Patch Set 5:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1179/ : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"a0cd697733890e6fb0914c3681ef3a07d58c0bc0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-19 15:52:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-67-977767-5-fc : SUCCESS in 53m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-67-977767-5-iscsi : SUCCESS in 1h 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8ade43f733d64398e0430bacb4b9a73f59670161","revisions":{"844457a2fc5d0a2503cde3ee9dd965ac5c28c5aa":{"kind":"REWORK","_number":1,"created":"2026-02-24 09:29:54.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/67/977767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/977767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/977767/1"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-18 06:12:10.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-24 09:29:41.000000000","tz":0},"subject":"Dell PowerFlex: Fix backend user rotation","message":"Dell PowerFlex: Fix backend user rotation\n\nPreviously, the PowerFlex backend user was stored in the\nvolume_attachment table and reused during detach operations.\nIf the backend user gets changed, detach operations failed\nbecause the driver continued using the previously stored user.\n\nThis change allows the backend user to be configured in\nconnector.conf. The driver now reads the user from this file\nduring both attach and detach operations, ensuring consistent\nbehavior even after the backend user is updated.\n\nCloses-Bug: #2142083\nChange-Id: Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/844457a2fc5d0a2503cde3ee9dd965ac5c28c5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/844457a2fc5d0a2503cde3ee9dd965ac5c28c5aa"}]},"branch":"refs/heads/master"},"5e412035166a1376f2d8d9d128e6b21bf135cf83":{"kind":"REWORK","_number":2,"created":"2026-03-12 15:12:53.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/67/977767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/977767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/977767/2"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-18 06:12:10.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-12 15:12:39.000000000","tz":0},"subject":"Dell PowerFlex: Fix backend user rotation","message":"Dell PowerFlex: Fix backend user rotation\n\nPreviously, the PowerFlex backend user was stored in the\nNova block_device_mapping table and reused during detach operations.\nIf the backend user gets changed, detach operations failed\nbecause the driver continued using the previously stored user.\n\nThis change allows the backend user to be configured in\nconnector.conf. The driver now reads the user from this file\nduring both attach and detach operations, ensuring consistent\nbehavior even after the backend user is updated.\n\nThis issue only affects legacy attached volumes. New volume\nattachment and detachment operations will not encounter this\nissue since the mapping now happens on the Cinder side.\n\nCloses-Bug: #2142083\nChange-Id: Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e412035166a1376f2d8d9d128e6b21bf135cf83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5e412035166a1376f2d8d9d128e6b21bf135cf83"}]},"branch":"refs/heads/master"},"1815fe700c3d540bf5c77a36ebc8d9004bb62ccb":{"kind":"REWORK","_number":3,"created":"2026-03-12 15:17:34.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/67/977767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/977767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/977767/3"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-18 06:12:10.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-12 15:17:27.000000000","tz":0},"subject":"Dell PowerFlex: Fix backend user rotation","message":"Dell PowerFlex: Fix backend user rotation\n\nPreviously, the PowerFlex backend user was stored in the\nNova block_device_mapping table and reused during detach operations.\nIf the backend user gets changed, detach operations failed\nbecause the driver continued using the previously stored user.\n\nThis change allows the backend user to be configured in\nconnector.conf. The driver now reads the user from this file\nduring both attach and detach operations, ensuring consistent\nbehavior even after the backend user is updated.\n\nThis issue only affects legacy attached volumes. New volume\nattachment and detachment operations will not encounter this\nissue since the mapping happens on the Cinder side.\n\nCloses-Bug: #2142083\nChange-Id: Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1815fe700c3d540bf5c77a36ebc8d9004bb62ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1815fe700c3d540bf5c77a36ebc8d9004bb62ccb"}]},"branch":"refs/heads/master"},"1f327d7da8c22f131ed66e289560a5b8540ea070":{"kind":"REWORK","_number":4,"created":"2026-03-16 05:12:46.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/67/977767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/977767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/977767/4"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-18 06:12:10.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-16 05:12:32.000000000","tz":0},"subject":"Dell PowerFlex: Fix backend user rotation","message":"Dell PowerFlex: Fix backend user rotation\n\nPreviously, the PowerFlex backend user was stored in the\nNova block_device_mapping table and reused during detach operations.\nIf the backend user gets changed, detach operations failed\nbecause the driver continued using the previously stored user.\n\nThis change allows the backend user to be configured in\nconnector.conf. The driver now reads the user from this file\nduring both attach and detach operations, ensuring consistent\nbehavior even after the backend user is updated.\n\nThis issue only affects legacy attached volumes. New volume\nattachment and detachment operations will not encounter this\nissue since the mapping happens on the Cinder side.\n\nCloses-Bug: #2142083\nChange-Id: Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f327d7da8c22f131ed66e289560a5b8540ea070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1f327d7da8c22f131ed66e289560a5b8540ea070"}]},"branch":"refs/heads/master"},"8ade43f733d64398e0430bacb4b9a73f59670161":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-03-16 05:15:32.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/67/977767/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/67/977767/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/67/977767/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/67/977767/5"}}},"commit":{"parents":[{"commit":"d17b5113512c7db12f9127a6ef768f9b2f865b65","subject":"Merge \"Require Gazpacho oslo.service\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d17b5113512c7db12f9127a6ef768f9b2f865b65"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-18 06:12:10.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-16 05:15:31.000000000","tz":0},"subject":"Dell PowerFlex: Fix san_username rotation","message":"Dell PowerFlex: Fix san_username rotation\n\nPreviously, the PowerFlex san_username was stored in the\nNova block_device_mapping table and reused during detach operations.\nIf the san_username gets changed, detach operations failed\nbecause the driver continued using the previously stored\nsan_username.\n\nThis change allows the san_username to be configured in\nconnector.conf. The driver now reads the san_username from this file\nduring both attach and detach operations, ensuring consistent\nbehavior even after the san_username is updated.\n\nThis issue only affects legacy attached volumes. New volume\nattachment and detachment operations will not encounter this\nissue since the mapping happens on the Cinder side.\n\nCloses-Bug: #2142083\nChange-Id: Icb0ae9fa7382e1a68cd9d78265258cdedd2343c0\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ade43f733d64398e0430bacb4b9a73f59670161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8ade43f733d64398e0430bacb4b9a73f59670161"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
