)]}'
{"id":"openstack%2Fcinder~977851","triplet_id":"openstack%2Fcinder~master~I6320b495e7da077fb7204b50c8d06d736a36e946","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-08-13 11:12:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-13 06:47:44.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-13 06:47:44.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-11 15:33:38.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-12 15:27:35.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}}},"hashtags":[],"change_id":"I6320b495e7da077fb7204b50c8d06d736a36e946","subject":"Run Cinder\u0027s Backup service in native threads","status":"MERGED","created":"2026-02-24 18:39:00.000000000","updated":"2026-08-13 11:12:24.000000000","submitted":"2026-08-13 06:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"977851-eventlet-removal","meta_rev_id":"11a0a50a6eea6dea8a5685aab33b32a005466aaa","_number":977851,"virtual_id_number":977851,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 06:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-08-12 16:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-08-12 18:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-12 15:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-08-12 18:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 18:47:30.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-02-24 18:49:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 19:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 19:17:02.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-02-24 19:21:22.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-02-24 19:21:54.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 19:25:42.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 19:25:53.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-02-24 19:27:27.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-02-24 20:39:50.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-02-25 00:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-25 03:18:33.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-02-27 20:04:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-26 21:02:55.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-04-06 12:43:44.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-01 18:55:09.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-10 13:14:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-11 06:07:03.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2026-06-19 11:23:00.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-06-19 19:55:19.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-06-22 13:13:39.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-24 11:32:56.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-29 14:42:58.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-28 20:16:59.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-28 20:17:12.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-05 22:53:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2026-08-06 22:05:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-12 16:19:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4be72268a24fe1367fed7703f2212b13339df384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-24 18:39:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f2a56bda36db16cf549bc6db410bda5663ba8d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-24 18:41:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59c02c88a5c0c2874b955888a44e69a6537487fb","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-24 18:47:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/cc62ca0b624f4a5985d8773abfe10718 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":2},{"id":"c7a9003eb8d64d571ad5f6e1974ab6f36d7bd1a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-24 18:49:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bb940d51605242b494d6f3bb769c0452 : RETRY_LIMIT in 1m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7b66100b192444a7bc1912f53e209093 : RETRY_LIMIT in 1m 06s","accounts_in_message":[],"_revision_number":2},{"id":"8df833d9bba909dca83475e43f928c1894d9de84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:13:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3809f5e35c40be82220345e883f361\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c641f0d09bcb4cbc889d1f447983f6ae : FAILURE in 5m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ea04a0966ad4a7995d975c1ffcc92c2 : FAILURE in 5m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5459f475b61649fb9da033e491ba1f72 : RETRY_LIMIT in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"243cfa44fceabd28495909a6eda02149af62c66d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-24 19:17:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/1a51b7d445164ac6a0e35e7569259077 : FAILURE in 33m 01s","accounts_in_message":[],"_revision_number":2},{"id":"374de2a2e6ed4f0191fadbf5fdab97bf0473a23d","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-24 19:21:22.000000000","message":"Patch Set 2:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/977851 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7e6962d8c2fe86c06f95300e567a83716dc95490","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-24 19:21:54.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":2},{"id":"7defc4413f7f6fe6857b9e81e2eae9861dd14065","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-24 19:25:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4208b619daea4469bfff306450966271 : FAILURE in 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d1fefdcafea50513d0f51860ef68bd46352b1354","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-24 19:25:53.000000000","message":"Patch Set 2:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977851 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"30e54314adc9aa0f3ef0b9838fee63fd71aed9b9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-24 19:27:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977851 : FAILURE in 39m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e763f8fd576d39b1afece85ccfe7468c5901f443","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-02-24 20:39:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bed43928e538e802a01cb4ad14c4c57a6783b89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-24 21:39:37.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b2f08541c41b91c9a19799638dced6af823beb65","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-24 21:49:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7f650f98525b4a539de15f23eccb5ec0 : RETRY_LIMIT in 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4984ea22fd854988a2bf24c8c009c321 : RETRY_LIMIT in 56s","accounts_in_message":[],"_revision_number":3},{"id":"5190a8e0b76975fc0b6bc8b8294dd2a3acd83995","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 22:08:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f60dc3360841888673b8eb988d4ee3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d382b6bed2274cf9958bb0351cf007ef : FAILURE in 8m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/90fd994a0f364e90829c18263031f940 : RETRY_LIMIT in 3m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2ab8abdda27a4f038f2697ae317c09b9 : FAILURE in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78cd6da9eb5aee0ef360c38cb9932f71e58cfc44","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-24 22:19:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":3},{"id":"d640d5487a8b32bcbe28c787e923ba2270e70a68","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-24 22:26:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/5236a4f85b8c43a4b1023edc49bd2256 : FAILURE in 42m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ee52059c1aa3fa708e48e84ea2c0a258f73cac87","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-24 22:42:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/e9b06e4b324b4dccb44fafb806bfcd53 : FAILURE in 32m 41s","accounts_in_message":[],"_revision_number":3},{"id":"8ba9c075583de9069e1eaf9ac14e935da20eba0d","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-24 22:48:18.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/977851 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bc2bfebd89de7cae6310e74ede4f47cd2f2510c3","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-24 22:53:03.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977851 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6981fa2e6cb0cf61136e25e7670baf88a9eea4eb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-24 23:03:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a642fdfe232047fdaccc634db5b2dc3c : FAILURE in 40m 43s","accounts_in_message":[],"_revision_number":3},{"id":"8aeba00bdf9d41c1bd7987701ab8aab472aab096","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-24 23:04:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977851 : FAILURE in 39m 31s","accounts_in_message":[],"_revision_number":3},{"id":"13a7f3ff9b0d1ce0347d9cd803da051c34c2072c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 00:28:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a158c3b6284b4e13a5283c9a84a32d6e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af591350e654d8f9d2c5b7602076a32 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12b69435117947f4b0f8eda8b9deee8c : FAILURE in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6582fd912c114d768514af5aba6c8046 : FAILURE in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76b835d821894a88bfc5a76be2b7175f : FAILURE in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f032918ad837461297b473bed5da966b : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0ff408835551400cb3e8724ecc40c522 : SUCCESS in 1h 05m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a46547fa32d41ab800038fe9d794a52 : SUCCESS in 1h 04m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4721e8deda4d4208b5da4f9640ad7de6 : SUCCESS in 1h 36m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e89d2c085f74dc99f55d7bc781f2170 : SUCCESS in 36m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/596141b62bf9497d8aef55fd1dd8f95f : TIMED_OUT in 43m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b32c159e404b4b6da36e190fc6bc6c12 : FAILURE in 6m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b289ef0857b5452b99dc2ade5a99cf51 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3b248bd5fe34707a2e24587168bd6db : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2edff68e63e419683b334b6953e9728 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fb36c53121844d4391a3cecce6f07f16 : FAILURE in 19m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d5945ee487841c1b435b696f5c4dbf6 : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a30a343a5434033b8d600999ad3e25f : SUCCESS in 1h 11m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/62f0eca8c0a64bae87a3b5e8da53602d : FAILURE in 2h 32m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bca642814d044f91981a7519599eafae : FAILURE in 2h 15m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/608c576b211d4a6a89cf0342b54784a2 : FAILURE in 19m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8dcc3ca9e8014cca85a468663d4128f1 : SUCCESS in 29m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/671d7b76152a46e59d7ec1b516396837 : SUCCESS in 36m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1985e62adda2421ca113da1fc1aee750 : SUCCESS in 1h 35m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2d4f8788ee8249cf9acea9434effde8d : SUCCESS in 39m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/13423691c5d64080a3df89d111c4a7d5 : SUCCESS in 1h 59m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9fa376ed17084dfe8b0d740a5b6297be : FAILURE in 19m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e4d94423e6e471584f1b65527bce385 : SUCCESS in 1h 23m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31130e725f154ff1a66286b62e42c0ed : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":3},{"id":"fba5b0b2f7f40c9adace2e551da37c352a2fabcb","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-25 03:18:33.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_842/index.html : SUCCESS in 85m","accounts_in_message":[],"_revision_number":1},{"id":"abe07259b363bdecf45fddc18ba0c194770ce809","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-25 04:53:37.000000000","message":"Patch Set 2:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_843/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":2},{"id":"641d0dea6e91e6d14f0d09f5f940ba27dcf018f2","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-25 08:17:25.000000000","message":"Patch Set 3:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_845/index.html : SUCCESS in 91m","accounts_in_message":[],"_revision_number":3},{"id":"39362c1821b73a7e3672128513aa2823981f5953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-27 20:02:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cf306d241eb50a6677b7402f6c6639c90ac6a867","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 20:04:42.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3103/.","accounts_in_message":[],"_revision_number":4},{"id":"e767e733d5f8c90b9ca817d9f1aa70ca61d81540","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-02-27 20:12:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c9cbcc91c01cdc6217c5bb7a509857e1cf39dcc6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 20:12:26.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3106/.","accounts_in_message":[],"_revision_number":5},{"id":"880696ac2e924b78a3fdd3c4b44228ea9590ca7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-02-27 21:02:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/10a93ba2574f49c6b1c7104273916c08 : RETRY_LIMIT in 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/cc5e0ed3051c4e68ba8abc0f52b1c4b8 : RETRY_LIMIT in 51s","accounts_in_message":[],"_revision_number":5},{"id":"b604e5a475f4b824e29f2e5b9f2054c6c9675a8c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-27 21:27:03.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":4},{"id":"0ae2bdcca8c45594afa37edf62d036ef42695190","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-27 21:33:26.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":5},{"id":"fef10896ee80f15c913b07ad52821c4f7d005317","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 21:40:29.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3267/.","accounts_in_message":[],"_revision_number":4},{"id":"614ea1fd120530eeb6935047351392bfeac45784","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 21:58:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97aeeb1c5d3c49edb925e5a782fe6ab1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12d905f63af34644af13866892194adc : FAILURE in 6m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7fb4c91e61c846deaa0c6f9b69bbe954 : RETRY_LIMIT in 4m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0e59ed5301f4b9a97fd24dfcc8d60b2 : FAILURE in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7a1f22699dc485851cd6b4c54b8ddaf66a9667ae","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-02-27 22:08:06.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3270/.","accounts_in_message":[],"_revision_number":5},{"id":"5da43cf5dda9e4691349c484339fd061898ba77c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 23:16:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e6ae8b5fa49475896c9a87e1b595bd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c6565db181405c9125e69ae7ef43f2 : SUCCESS in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/200e3fee4f8d4a0ab2eb29f620481caa : FAILURE in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e96e2fa96f54d2bacbc1bb08acf3610 : FAILURE in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c3bbdb04dfc4c49af15a6aeb331c53e : FAILURE in 16m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8214224be4174fd084d09cf79c9bc89d : SUCCESS in 18m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f141bda5d8914582a48a6333e99a651f : SUCCESS in 1h 04m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28064133febb405ea0f95c36b06e6d5a : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a63c26bd417441719e817306a1c2e082 : SUCCESS in 1h 44m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc1d4af39c88483d8e3dad8a50e3959b : SUCCESS in 59m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/38b81a058498412d8f3addc6c48f002c : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8c321e317a4e4c9ba4999fc8ab1f3676 : FAILURE in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/313a42d220e64ae68895f26efcc1008d : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c6c9c5742d541ac86fbfb6554708e8d : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3a9ba113a1c448ff8bcc7fb987f16fd5 : SUCCESS in 9m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b68a130f692d4afab717c1fbf9573d65 : FAILURE in 43m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10be7f15049548a886303d0d70f10be4 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbc3cf9edd1143b9b5d77dc13cc8cd01 : SUCCESS in 1h 56m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fef83905aeec4768a95be30348e5065f : FAILURE in 2h 45m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02c3ac70ab1d419f8bf913277a914ad6 : FAILURE in 2h 20m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/de238cfe97e0488aa966fafa403e0595 : FAILURE in 19m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5e389dfde1954c25be287f7196cbdf51 : SUCCESS in 41m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8f5f1b4873424f90932968c6fd817a25 : SUCCESS in 48m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a21304881155404d9f7ac39d994ddf5f : SUCCESS in 1h 40m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd460b7dbec44d6a94ace92a10d7c1be : SUCCESS in 44m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d1179fd546da400e88b69b6d0c3b9516 : SUCCESS in 2h 00m 56s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/b2112100574c40e9adfd564befe5bfc9 : FAILURE in 17m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/efaaf74325504107a50b1bf4747d50bf : SUCCESS in 2h 03m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f2514a9fce5443f91344a3e2da808b8 : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":5},{"id":"16ce7aeda3d1089997dadc54df3d8bd04dd59121","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-02-28 01:09:10.000000000","message":"Patch Set 5:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/977851 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"6a72029cc10e32c48ed72c2debb6fde44e62e37b","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-02-28 01:09:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/b77c4357811b407a8d2c679ce37a707f : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":5},{"id":"208368ecb5df70b894a9c4f730d4eb1cee717195","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-02-28 01:37:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ca5bfcd45dda4f52b9662cd023d77ace : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"e6ad5095ac96bb37bc80cbcf0f7a2b4fba73b9d3","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-02-28 01:39:44.000000000","message":"Patch Set 5:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/977851 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"0dfacf5ef31c0891a4e9cf0b87bf5887b290bf3b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-02-28 02:29:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977851 : FAILURE in 11m 42s","accounts_in_message":[],"_revision_number":5},{"id":"eeedb4c334c481ff41a23e84f3d915e53818dbac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-02-28 02:51:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/83d56823ab9f4769a9879c48551b9f65 : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":5},{"id":"e853647afdb0325ad548e89e6ff57df8b63f82a2","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-02-28 17:43:04.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_899/index.html : SUCCESS in 92m","accounts_in_message":[],"_revision_number":4},{"id":"a25986f74dcf41ff43d5cab13e6335ff6c914fa5","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-01 04:47:21.000000000","message":"Patch Set 5:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_901/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":5},{"id":"14eca0e3d980671f68838ca164196d1488889b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-26 17:53:37.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"353a5b3b29817e51d869a69a5535ffc8941b9890","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 17:59:59.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/737b7675416a414d857b72f13612010b : NODE_FAILURE Node request 200-0000679119 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"54c03ef0abafa2e391dbcc0ea66f4e50f7d9bfba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 18:22:25.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3f241794a10e46ebb6c849e9dab0320f : RETRY_LIMIT in 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/074b36b028f949f0855fd2d10544334e : RETRY_LIMIT in 46s","accounts_in_message":[],"_revision_number":6},{"id":"89422615a8db7e14676531851c51ce9d68702d44","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 18:35:20.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":6},{"id":"e5143e7152849d40689a956d4f4545003bde0324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-26 18:46:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7066762a15aace90357c166933db7bdd4ee60a0d","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 18:49:04.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3dd1cb181ffc4ba58352cfd592a08769 : NODE_FAILURE Node request 200-0000679221 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"19292a1aeb930ce9346af651fd9456588f71a706","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 18:58:12.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6fd8d431b62a47e7a14dd8cc0b31d517 : RETRY_LIMIT in 46s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/fd62d0ed2b814d0e8b5455fb02539009 : RETRY_LIMIT in 46s","accounts_in_message":[],"_revision_number":7},{"id":"bbeace90d5d0c872e134d5860b3a0152999d9adb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 19:23:36.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":7},{"id":"f745f78095a8a4c6e4cb7c27ad409cbbee8f64a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-26 19:39:15.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"662f1a23f2516eb6aa61099302f850d9f27355cd","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 19:44:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/83e9d68c60424678a176c364ff236579 : NODE_FAILURE Node request 200-0000679335 failed in 0s","accounts_in_message":[],"_revision_number":8},{"id":"f5c78506a9fc23481522527d3b092bcfb5383cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-26 19:54:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d19cded8c9c5548975ab8e44814a119f505cccee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-03-26 19:56:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c8ef8d366db8244ba6aa9a8d28a1b6fa3b7c0094","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 19:59:31.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/2f70ceef3285477dbf74407ad0acb681 : NODE_FAILURE Node request 200-0000679370 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"768015c02eac9e80a681e3554d554e53c905d5a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 20:10:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/20bce2f52eac485a918c669d33ddbf84 : RETRY_LIMIT in 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1d570f8155b947029fc6577a76a2090d : RETRY_LIMIT in 45s","accounts_in_message":[],"_revision_number":10},{"id":"689602fdd6144c4e48269cbf630f830f76ceb352","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 20:19:27.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":8},{"id":"6820fec061520178aa450419416612e708c73899","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 20:35:25.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":9},{"id":"c346e53dcb6136b6686169e67d5788c77c84de5e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 20:57:11.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":10},{"id":"8d32f7a0cb9d34a819f1863d91d16e26b7942986","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-26 21:02:55.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-6/FC/ : UNSTABLE in 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-6/iSCSI/ : FAILURE in 3h 8m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"157d2cad2f071aaf27337d547d298cef7c73428a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-26 21:40:19.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/981768, https://review.opendev.org/970168, https://review.opendev.org/981407, https://review.opendev.org/981107, https://review.opendev.org/959385, https://review.opendev.org/970332 are needed.","accounts_in_message":[],"_revision_number":10},{"id":"704e3af4e8bd670993e36d8e314716a229e7648e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-26 22:26:42.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977851 : FAILURE in 38m 39s","accounts_in_message":[],"_revision_number":10},{"id":"4d198365a7a8a314cd49f075e1ffb5d6bbde54c2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-26 22:33:07.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-7/FC/ : UNSTABLE in 4m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-7/iSCSI/ : FAILURE in 3h 46m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"bd1222dd3643a7f2634d61b8a33d2aef79eca584","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-26 22:37:08.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4bcec37395e446d0abb6cd1a623b923b : FAILURE in 38m 31s","accounts_in_message":[],"_revision_number":10},{"id":"185c7d3219498e92cd6b9e5b677b5ccaed5aebdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 22:39:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4c4f335441429abf79749ca2caca57\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e4225a1fddfd4e4ba652313340656438 : FAILURE in 36m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c8c51528db048fcb3f1961866fc472a : FAILURE in 33m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b99113860d304d2d922f7ec105b4aa8c : FAILURE in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6fc09dd2af685ecad2e704952d551133cc5a8556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 22:53:28.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b100594994449988f374f9b3dc636f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5da75e6aa10344a2b48913901c7d3a50 : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c068c70cf269445c8db361d955527cc3 : FAILURE in 33m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cabf41fdfaa54cafbba21b91519689d5 : FAILURE in 28m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fbe46c2d0c44e5dbcb4dceea142915c : FAILURE in 36m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a77f4af9ee34e6ca6520daa5594969f : SUCCESS in 14m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c2b6bb88dee34ed4961253cbe77c0d2f : SUCCESS in 1h 10m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a9f0702b245461d808c1901bd1063dd : SUCCESS in 56m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f10a347075e4cb3a55d999c3a6cb623 : SUCCESS in 1h 47m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc70e3b021df484b9bd90671d65d7694 : SUCCESS in 57m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1cf0fe65475547d68d34c6144ec35bd3 : TIMED_OUT in 41m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eefb057bcbd14f9b8dafcf578dedd3da : FAILURE in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2e15784cd68747ba83558f8fa41737cd : SUCCESS in 7m 27s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a32b47b5189f42a2a80c911f65b9e206 : FAILURE in 34m 16s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36c7b773a9cc4c8da24ba97dd9ecf492 : SUCCESS in 8m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4d7820da6b8c46fbb6008afad124a5b7 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af313c870b9f4ca0944c0ec1358792e2 : FAILURE in 34m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e734594e27b1492cbec428f5ffb19e5b : SUCCESS in 8m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd372b718d4246d4b9e6ff02df481d94 : SUCCESS in 1h 55m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/128ffbfcc1a14afa8557789b638b365b : FAILURE in 2h 49m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/72b9049f2695460c97cd43c7fbc3186d : SUCCESS in 1h 53m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0dc89c9bc4a54d14a31bd951e3f410ea : SUCCESS in 1h 53m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/005b620e7a1348a485c5c476a40787c3 : SUCCESS in 37m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3e3c5a0380d740c5a23f0f0f3a746614 : SUCCESS in 1h 37m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ed4522fc11fe44cb87fc45324824cd60 : SUCCESS in 1h 36m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e23a1ddce6ff4b898b1c4bd9703b9851 : SUCCESS in 40m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f6d1c3c45923416aa2165e12ee576893 : SUCCESS in 2h 13m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/30097aad19604d6d8af4044adf5fa5ea : SUCCESS in 2h 16m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b63de79d79644676aa9f9685a489383b : SUCCESS in 47m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8bee55c6976f46e28f1dbcc4c004bde2 : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":10},{"id":"507c4ea9ca83411d420a2416aeed5bc9fa68af3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 01:28:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3596/.","accounts_in_message":[],"_revision_number":6},{"id":"69fc38f512888d9b3b7188db57b5ef83ab8178b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 02:39:18.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3597/.","accounts_in_message":[],"_revision_number":7},{"id":"5a6a3546190cbe9a52a5047ebedc33bc073af67c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 08:22:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3602/.","accounts_in_message":[],"_revision_number":8},{"id":"2679f06e075852b4105ecab97444f739bcedd261","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 10:52:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3604/.","accounts_in_message":[],"_revision_number":9},{"id":"57fea51563be60b84a36fd2162d0d3b52c94105a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 12:58:14.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3606/.","accounts_in_message":[],"_revision_number":10},{"id":"06096379f08cacab6b548a8b53da4c2cd7c2476b","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-27 14:59:40.000000000","message":"Patch Set 6:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1247/index.html : SUCCESS in 90m","accounts_in_message":[],"_revision_number":6},{"id":"43a46455d2490b6416c014dcee6e40b6b3a87527","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-27 18:31:20.000000000","message":"Patch Set 7:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1248/index.html : SUCCESS in 110m","accounts_in_message":[],"_revision_number":7},{"id":"648d53be6f67b192bcc4af63124dd23db2d85861","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-27 23:24:00.000000000","message":"Patch Set 8:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1253/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":8},{"id":"c4b4cd6c71c1731c5e69b0f2e32dd194953d82b3","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-28 04:29:37.000000000","message":"Patch Set 9:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1255/index.html : SUCCESS in 104m","accounts_in_message":[],"_revision_number":9},{"id":"04a3336e0928c59a17d2aef90dc2c9ea9c9afd5e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-28 07:36:21.000000000","message":"Patch Set 10:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1258/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":10},{"id":"62a739fae81b3f23000a5edbc8608146e339be4c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 19:03:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3423/.","accounts_in_message":[],"_revision_number":6},{"id":"597ccf9e9a711fb617ad10dd501da27a0d9de1b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 20:06:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3424/.","accounts_in_message":[],"_revision_number":7},{"id":"1b1f2a76c2dee8cbb34d76f56210a2e37bed923c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 01:47:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3429/.","accounts_in_message":[],"_revision_number":8},{"id":"21fd0391072cf2a287e655ab3ede8e8657d5b3b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 03:59:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3431/.","accounts_in_message":[],"_revision_number":9},{"id":"04ed760b5318cfccc76d716c4c2835945b27aa21","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 06:32:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3433/.","accounts_in_message":[],"_revision_number":10},{"id":"fd74d3add261026f6efd901884d81d3d38944021","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-02 14:44:45.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"94a2af0cb05df8742cd8556219ea3e96941d5976","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-06 12:43:44.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5ad2a35fff8179c3a7966abf226b065f0c762cc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-04-16 15:04:15.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"68610e36d621ca68421477eb01cd7122b039b426","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 15:26:12.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/712e919479e4482ca634934f65da18f7 : RETRY_LIMIT in 52s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a596cf149aae47feb4e6dafb5d03a08d : RETRY_LIMIT in 49s","accounts_in_message":[],"_revision_number":11},{"id":"2ec2573c3298addcc977ac445352e6737b2d3127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:47:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b378f9b8a3e45ff8522ae3d0091c0d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38701ce450444c7d96f8897a8b7d2084 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3c9fd40b7464e27a9c434761cbcd30e : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b8235ad05bf409f9e8ce5fb2291b58a : FAILURE in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c3f7e86061e14c4fc8ccfacac085fd93d75a89d4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 17:49:25.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":11},{"id":"5e15609db9413923aa2bd56097b369510b124879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-04-16 18:28:13.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"096857cb66b3fae338ee63b3e8dbc7334200edbf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 19:03:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0dd0f2da98614c71a71787db4198a0da : RETRY_LIMIT in 49s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/98ea6edf1b8a4332b3503d6000309202 : RETRY_LIMIT in 48s","accounts_in_message":[],"_revision_number":12},{"id":"589cac23a4921427736a2e9d2ec73aed3e968e74","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 19:11:18.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":12},{"id":"15c5bb1fe41bc327efdede9828e0366d20430d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-04-16 19:44:03.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"ee70dfa1937c8ddb0569031ecee834277b06bf14","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 20:20:09.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d4a0060f2a024771b1158aedd585d6c4 : RETRY_LIMIT in 48s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/006dc1db0f47494ab5e99e39ee34136e : RETRY_LIMIT in 52s","accounts_in_message":[],"_revision_number":13},{"id":"4f42faddb8a5b8618c8d062331d881e61f11d89d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 20:28:02.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":13},{"id":"bb8f9e9d1216af33bb6ee76cf3b3b39e7c1aa1fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 20:37:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96e25e385ff94e69810166dc6061a922\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18232a086bc7495a9d091edaf266ea3a : FAILURE in 7m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/339660da76f24e68a3457a38d1fd4359 : FAILURE in 7m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b77c6ba80f140469c277ba4565d1ae1 : FAILURE in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"64429d9ac9d62c47c8268a75b81e890a90eee1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 22:30:33.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/748ca2a4211a460aa9b5fd22a74892ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcaf64095f3b41fb9ad16b72749146b5 : FAILURE in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59573d28f3ad4bf7acfc50053fb748c5 : FAILURE in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/990f75960b03401b9ca813a8a03c89d9 : FAILURE in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62c23e6530f04929b9cdc0ea73907d9a : FAILURE in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da526cedaf143ad8056ae6271fe2161 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8a854762d80e441b9f7cec1c67d2e9cc : SUCCESS in 49m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6cacd64a77fb414db97f1a272cb9b235 : SUCCESS in 1h 21m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d52c3dc50cf94318987ba5d7623b00ee : SUCCESS in 1h 52m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c38bf870dce478396de09a8925bea36 : SUCCESS in 1h 03m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/538e33f1b00c49f7a800e5a95f78f239 : FAILURE in 6m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e5bbc0c72394961b9d16c2628cd8c28 : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4cd88592807445b1bcf8d26dc13c32d8 : SUCCESS in 8m 07s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f6e8bdac07d1477198797d3aa86b9c3a : FAILURE in 5m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eddabce2243f4834826c91b0a05c71fb : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e6df4cf94ee149649ec785a5220d4168 : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a8026e30d3f7487f90fdc7e44b05eaff : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1551092a85ad4a0997eef222da182700 : SUCCESS in 7m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/538612fab5b542d0a0fdf7fbfa21870f : SUCCESS in 1h 52m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01f2695bd78a4133a86069018499b5eb : FAILURE in 2h 31m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e001f8cfaf349b3b1ecb3c535008589 : SUCCESS in 2h 02m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/90f83b33bf114926abadf6111c5f4b50 : SUCCESS in 2h 08m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b219fa64058e436687439c2c85f90e89 : SUCCESS in 36m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9598ba423bd247318cf3b918f5b8434f : SUCCESS in 1h 17m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29237be042cc407e958fba41ae884516 : SUCCESS in 1h 55m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a09e9a9f07ea48d1bf12715154d4e82b : SUCCESS in 52m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc8d0a970a254bb3803668bad4b32e21 : SUCCESS in 1h 55m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e35ecb13cd124547b047dcf79df8e9ec : SUCCESS in 2h 06m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/359299e4eb11418ea29c46dc3e397dee : SUCCESS in 33m 10s\n\nWarning:\n  Comments left for invalid file cinder/volume/drivers/rbd.py","accounts_in_message":[],"_revision_number":13},{"id":"96233bef35180a12816af96adc99aa74b9fb8f2f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-16 23:40:41.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/fbd3706fa4524d9493673b6218c09fbf : FAILURE in 24m 21s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851/etc/tempest_log.log","accounts_in_message":[],"_revision_number":13},{"id":"a15e8f783bbbe5df192824c5ff355c0d54a4a96e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-17 00:24:32.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- hpe-fc-driver build/64f390690cc34206a92b0254b1cc98d1 : SUCCESS in 37m 34s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851/etc/tempest_log.log","accounts_in_message":[],"_revision_number":13},{"id":"af0d496dccf0b7b4be709a11986d43b7402aaa21","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-17 01:52:45.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/977851 : FAILURE in 31m 26s","accounts_in_message":[],"_revision_number":13},{"id":"5e2795ff8310fc7fbe8b2182cbc5f6345c4d6266","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-17 04:04:53.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/080f496c0f7f4335a65c23a9380b5395 : FAILURE in 46m 44s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851/etc/tempest_log.log","accounts_in_message":[],"_revision_number":13},{"id":"b5fa380764c804e4e02d52cde301e46e055d9f62","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 14:25:38.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3677/.","accounts_in_message":[],"_revision_number":11},{"id":"8b88da03b0232d2513ffabda0fca7054d47c33fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 21:29:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3684/.","accounts_in_message":[],"_revision_number":12},{"id":"fdbe2923f4b2af7ca36ac640b3a898ffa6495445","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-18 02:29:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3853/.","accounts_in_message":[],"_revision_number":11},{"id":"bfbbb7c44e8c50e7281871f17e11d6082f344c42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-01 18:16:28.000000000","message":"Patch Set 14: Patch Set 13 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":14},{"id":"79707c3b64d8208351c877516f951ae220f7f7bc","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-01 18:40:59.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/00dbe19ed54140d285747ae2445b18ed : FAILURE in 22m 32s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":14},{"id":"e79124f7508acf7c7538ff15af04ae20e1c7c468","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-01 18:55:09.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- hpe-fc-driver build/fe442456071a4bc299820ad08fc8ac76 : SUCCESS in 36m 20s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851","accounts_in_message":[],"_revision_number":14},{"id":"cd4b8cc78db194491e45d716f8a72bdba766c673","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-01 18:57:53.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f06d99b3c0604a52828a6d6f04be2bb7 : SUCCESS in 40m 23s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851","accounts_in_message":[],"_revision_number":14},{"id":"6720b04cf9b6bb33942aad4188662ea69e698b5c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-01 19:13:25.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":14},{"id":"822019f24ead4bf57ea3704f05c240929e534d24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-01 21:54:51.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4313/.","accounts_in_message":[],"_revision_number":14},{"id":"4667d5bad6a7576b358e6bd732aafd4141716276","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-01 22:10:01.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4137/.","accounts_in_message":[],"_revision_number":14},{"id":"9682379366e19cfce399208d81b3d9c6b6423fec","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-04 14:58:31.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"50ac3ea30fec5b6375ba44b52d5f9c6c065b3e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-10 12:50:16.000000000","message":"Patch Set 15: Patch Set 14 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":15},{"id":"f4ffb95c9c7943a5352ca62f7e0de9e68858c05d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-10 13:14:34.000000000","message":"Patch Set 15:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/09a33a44eeb8419ca6800cad413288c6\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/85b6cd425d39471382779cddf380bac6 : FAILURE in 21m 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1754d49b6add41c1bb969a18ea638f7f : FAILURE in 23m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7428852c00eb42638d72e351efc55164 : FAILURE in 23m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6fec07c81faf41de9fb098bba572ebda : FAILURE in 20m 38s","accounts_in_message":[],"_revision_number":15},{"id":"e05d6b98ddced9d25126ff23a9d21e6e97abaaa3","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-10 13:31:15.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- hpe-fc-driver build/59fa0d76ae234592b1a4ecd3d649745b : SUCCESS in 38m 50s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851","accounts_in_message":[],"_revision_number":15},{"id":"c089a8b0204f338332c61393dcde76e9edff7658","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-10 13:38:16.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/363fe1194e8442548ac2dbfe0d4cf42d : SUCCESS in 47m 17s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851","accounts_in_message":[],"_revision_number":15},{"id":"e7baf610b0bb282d1f3635abad5e847a44205713","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-10 13:39:30.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/e5d99821872144e395b9623713db1448 : SUCCESS in 47m 28s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":15},{"id":"fe536e95540a88813a00f991e6d1ac41b52a6063","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-10 13:50:38.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":15},{"id":"11ff7ac982d176022c438784eb0d3788be584acf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-10 13:52:47.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e8c36ce07f32449f99fe84d917e25368 : SUCCESS in 1h 01m 05s\n- cinder-storpool-tempest-iscsi-multipath build/93f74ba11520484fb0077d009f425770 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":15},{"id":"22135fdf70156bfa3076ee4b5da3b709d752d650","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-10 13:55:15.000000000","message":"Patch Set 15:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4418/.","accounts_in_message":[],"_revision_number":15},{"id":"bf4c54c6729742944e85c212a0218b071971a273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:20:20.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfdd4f16857040769f4814ffaadfff1d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8f8791664074706948b059d21de778d : FAILURE in 6m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a024330829cf45c59f417446147c9740 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a35043a721d045f6959e025cb2aa2db0 : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"54618c5925485d23857f1130efceff5f687dfbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-10 15:30:58.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":16},{"id":"1087216626b790430db86c82c8a4dabb09ec006e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-10 15:31:33.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ae519c96843fbeaf83d22c493dceb1502a67933b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-10 15:54:58.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1391f4e2735545cc8329a4b7fed8a575\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3802159ddf84dfdb48f88c3bd544b49 : FAILURE in 21m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13ed23974d724368ba51f6a72a611410 : FAILURE in 23m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fe6d166e357494895cace3ac8d30f21 : FAILURE in 20m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7c579129c4c4ea7a4e0afdcc4ba6537 : FAILURE in 20m 41s","accounts_in_message":[],"_revision_number":16},{"id":"69b8ce0eeec16f60ad8510e13e17d3cb5965ab9d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-10 16:32:14.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":16},{"id":"48e1007f47c90d560ca50baa3262a7bb3fba15b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:14:42.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b7169f598e7444683e3c5b013d0e83a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8de76597507745d382e30fec32dca06c : NODE_FAILURE Node(set) request 1cb805fae06a44b1bbb30c8e2195632c failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5fedfff789374f128f0b714e7fce6ca5 : NODE_FAILURE Node(set) request fe557067123742afb1efc08696e97805 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/574bddbe3f3d4d16a4aeb4fec926556e : NODE_FAILURE Node(set) request 64d8cf3f47724966835e0ba065272d44 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"de0aee73f59602b6c81c5be14e434c8d38ca7159","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-10 17:21:36.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-fc-driver build/d6322d76e82e4a1ca9407f37550503a1 : SUCCESS in 40m 21s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851","accounts_in_message":[],"_revision_number":16},{"id":"cc21601ae48935cb420b3fdcc5f6b67f992cc46b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-10 17:47:44.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/2ff4ea9740d648f0a7838e8899498b9f : SUCCESS in 46m 33s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":16},{"id":"1d45549a888f332113b937479a8254189ed9864e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-10 17:51:14.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/87b7a2bea74e4e44869588e6872c47d3 : SUCCESS in 40m 56s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851","accounts_in_message":[],"_revision_number":16},{"id":"e93687c096408a19d47f74fc58ad06cb10a1091b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 18:44:11.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26860c3d05474e93b1d62f40290291a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a6afb8c4934341bb2813288851a15d : FAILURE in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/858acc13e6e94895af06ed029a59c600 : FAILURE in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8bd3e7212ac4ecca6292eb3099e1b05 : FAILURE in 3m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00b8c63d7c604d368def03dfd757028e : FAILURE in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf8abb15f104fb1b721945296339786 : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/68d2a998bd4a45d5a2309a251a38881d : SUCCESS in 1h 06m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1aaafcebc18848569f4042fa1f3d7e59 : SUCCESS in 55m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a02dc04f55c945978124d139c42b396b : SUCCESS in 1h 51m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/137e1ed99b774208b3eb350a812d818c : SUCCESS in 1h 04m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f03e522d922443008d8b103881609fe1 : SUCCESS in 9m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3c7f7dcd57dc4fed9f11737463aaf583 : FAILURE in 7m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e14b7cc12aa345c7a20d2ca08b0f38dd : SUCCESS in 6m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aeabc0fea99f4fa68e9f70248ff5c314 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/12d4cf48899b460f976d39091f3e0de7 : SUCCESS in 6m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5d4c64bc6b04493fae660ab70b0f2e22 : SUCCESS in 6m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9980194e10ee4aeda630b145d669fbf2 : SUCCESS in 9m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3fb82e3e0d8d4e64a47d1fc2ab4cdf82 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58bae58516ac4bad88fd5eb87f48e144 : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a95c422d4a3842b481c30bf09f226e2a : SUCCESS in 1h 05m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/451d2e0be46c4ef7bb5b8f839eb5d6bc : FAILURE in 3h 00m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e79003c8c0e74399bbfcad473eb7962f : SUCCESS in 2h 04m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0ef2cce792ab4672a73f24ec0c67e75f : SUCCESS in 1h 53m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9c3f56d580cd43b493cf935dd68c5565 : SUCCESS in 42m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/336af8cfb3dd4c0c95221a753caff6ff : SUCCESS in 1h 20m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca9c3209dc0045e9b2ecfabe96cb5ac3 : SUCCESS in 1h 43m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e18ecb16e4549aa9d6de415d09b93aa : SUCCESS in 42m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7aed47c74e3f46af9381e7d01702b524 : FAILURE in 1h 04m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b4f8eebbaa34672af33563145ab4811 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":16},{"id":"31ea71b355c4008e47fcfee04d7e9b84e60b701c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-10 19:32:53.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/901a46076da6434c9bc477168ecddd25 : SUCCESS in 1h 00m 57s\n- cinder-storpool-tempest-iscsi-multipath build/526eafaef59c4ea8b452f9d9f7a80bd6 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":16},{"id":"84d1afd5791c580fcdb1c1e4e93b72b10b7ddd3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-10 21:00:30.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4424/.","accounts_in_message":[],"_revision_number":16},{"id":"693b7115ab5d9e74ea401ddb9a631bf84f5bc11f","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 06:07:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e1d139c7f9a002ae0e1a8c3f111eb606b508aae7","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 06:07:48.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"32c9acb62d15e38c90851b4834022fbf00693919","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 07:03:11.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"ce400e253e56d545a56fcb3dadac6af9db59b009","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 12:53:51.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"cda025ea9a173b31d40fc29da7145af38ef76a64","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-19 11:23:00.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"7e105c58cfc379e76fb4bc7d89fbc4358b38f97d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:19.000000000","message":"Patch Set 11:\n\nBuild Failed","accounts_in_message":[],"_revision_number":11},{"id":"07174daf88fcd59a3fd05930eb425ee3441eb346","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:06.000000000","message":"Patch Set 13:\n\nBuild Failed","accounts_in_message":[],"_revision_number":13},{"id":"636937751b6d7a019a2574c92296b497143ebd07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:03.000000000","message":"Patch Set 12:\n\nBuild Failed","accounts_in_message":[],"_revision_number":12},{"id":"dd485bd2b226bb52b26e27f0c3726567f197f160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-22 12:29:25.000000000","message":"Patch Set 17: Patch Set 16 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":17},{"id":"764c77d4a4aad1566a4b1a81b318f311c649f144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:40:39.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee6d10ace014e8ebb7d6a5c590d62b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/719df16e6fe24a8e96a4f827e72cc804 : FAILURE in 8m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/19e1d7c767a54f9a981a6d9bb42625a0 : FAILURE in 8m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/39d4ed2e9aca414581702c7026480613 : FAILURE in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"996d9f206159135c9240d37339f371227107db75","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-22 13:07:53.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- hpe-fc-driver build/af89382a949743f58d001309dcac04a0 : SUCCESS in 37m 54s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"36462d7e6badd30877cd6144c4ee6f5c70ed8cb0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-22 13:10:56.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/5969174dde31425cadc9b5aeecb4b881 : SUCCESS in 40m 10s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"7d4aa0712436415f089190c04ae94ce38a64b13b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 13:13:39.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/265 : SUCCESS in 0h 42m 19s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"4b90f5c6cdbb50ea073ae9626a0a41e94a4055b0","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-22 13:17:59.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/ce453767bf6147498b2091551d14a6ab : SUCCESS in 47m 25s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"c457d923c58fe764416389838383720fdea56541","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-22 13:28:25.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":17},{"id":"eef41e08be32f738b92eea9e57b009591479e6ae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-22 14:33:22.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest build/70c1074db9e44068b46fc2aff786f63f : FAILURE in 34m 47s\n- cinder-storpool-tempest-iscsi-multipath build/aa137a48c8fd40919fc16588378456e8 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":17},{"id":"f61053b3acd4c56a84b62da4f585a45b2fa43611","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-22 14:34:17.000000000","message":"Patch Set 17:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0ccb21ac6c834fef90271999519ea660\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d440facdf2e43f0be39fc55a3525d7a : FAILURE in 34m 59s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ceb2a3ca7af4404a9d8a053428e89167 : FAILURE in 1h 33m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4c7cbc812e3c4669a9e067309ce17f28 : FAILURE in 2h 04m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bb1fb8c69acd4397885aa7e9ec988c33 : FAILURE in 35m 29s","accounts_in_message":[],"_revision_number":17},{"id":"673db85dcd4a949106e96965c11658392455fd1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 14:46:31.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5ba22e382e64c7393a56f097d53ecbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4314048a7d8742d5818d174c087442ae : FAILURE in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/013db6b33eb8489fb83430618b1e916d : FAILURE in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af74257b01e74018afc4b69e093c2f62 : FAILURE in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/511279e010cb4df992951e0f8d141349 : FAILURE in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab52682e45374d9f92c24ed11cd161db : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ad54aa6f21e84f88ba1d9dbe8864a5aa : SUCCESS in 1h 15m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e96d321e9dd44e60addf1e6ff1319133 : SUCCESS in 1h 09m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56eec5e2918e4b0594cee79453b74d27 : SUCCESS in 1h 19m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/869a17e2a631434b91c9401b5abf4ff5 : SUCCESS in 1h 04m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b511779c36945ef8b76128659b97851 : SUCCESS in 9m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e873dfbb6d0a40a2a081e7f649037b7f : FAILURE in 3m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e04e3a8f69247a2a75ab9d5914e606a : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/abe07e5e584a4c2aae86cb43f274b037 : SUCCESS in 6m 47s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/af03a99a3b354160b5b10d85a9a512a8 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbc8ce4d36bb4ce6be8be205ea9d98e3 : SUCCESS in 5m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2a7b39fa75a249eb929331dfe801bdca : SUCCESS in 7m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9e74266be26040cd93de5b039954de85 : SUCCESS in 6m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9446bb6c5a56419dbc9b33356ce7ecfe : SUCCESS in 39m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a227a731b314dc0b591d072dedc2f0e : SUCCESS in 3m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d06354bcbbf4983bead5b02f09a377e : SUCCESS in 2h 00m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e715ebeec5284e17993dd4fed0903922 : FAILURE in 1h 26m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7dbcd47f57a40f68d205c471f586d0b : SUCCESS in 2h 10m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/10b3790b1ca7461dacfcf9cd50b3ee3a : SUCCESS in 1h 56m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3382d119e70f442980a6eaf8af5d929a : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f0499407021046458698277a62ea9b54 : SUCCESS in 1h 20m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94a03d74c9104a25857dc2475399b713 : SUCCESS in 1h 43m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c51ed48d7fed49849f2a9e8e96e70589 : SUCCESS in 46m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5ce1735c35f43549c277914e4fae0b2 : SUCCESS in 1h 51m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f17b2a6e7ce461194d170af653da0fe : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":17},{"id":"ac6f00e351d0056421d7c644bee99c5b635d6ec3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 15:09:05.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/540 : FAILURE in 2h 39m 12s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"ae47d9bfc0f4ee89b8246b895e4471fcdd543bd0","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 15:10:31.000000000","message":"Patch Set 17:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/179 : FAILURE in 2h 40m 38s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":17},{"id":"2f6b6fa53a395dc4f90ac838acc132ce3575de7e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-22 17:36:19.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4472/.","accounts_in_message":[],"_revision_number":17},{"id":"941954a437658bcc66bb7926cc0f90ccb47c20f0","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-23 05:44:25.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"79305470f2bc8ee6a55fe72b52f1a9d8ef2aa8a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-23 12:38:22.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"91eb33f1673b288c7ef9af37889a9c00714c4554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:02:06.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/928c09f76ae748a3aaeab4adad26128e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd6ecfa49e3b48b0bd09c468f793b56e : FAILURE in 17m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/721903b573eb423f866e56b6ce6ca298 : FAILURE in 11m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a35d8bb3b464e8f8688e8b6ea1a897a : FAILURE in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1fb4be53b18d649abfd05bc36099e54afb0a2cef","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-23 13:02:47.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/290 : FAILURE in 0h 13m 2s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"74e9ae260fce7d0672ce4dc4af5a525a4f323248","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 13:40:30.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":18},{"id":"998a3c4c22e8270934074aa79b0dee8a1ae01cd8","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-23 13:54:29.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/42a18ba6fa4e40a4aa5d352547016e86 : FAILURE in 22m 52s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"3fac87830e6595fe009f0830cd83f75e0822ac94","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-23 14:13:11.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- hpe-fc-driver build/297d1ed4182a4fe7bfdd0a12cbb8c075 : SUCCESS in 37m 04s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"b1dcae15923a64adb0d2240e0f1dea6104e57620","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-23 14:20:22.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/291 : FAILURE in 0h 12m 45s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"186de449395eae225099d2ef53a12fc65399e66c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-23 14:22:19.000000000","message":"Patch Set 18:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/aa6cd7d699564ca5ab6267556d53c142\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4cbd7864f31449e5bc8150b091ff68f6 : FAILURE in 40m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7461731a0d254567a3752031e41f40eb : FAILURE in 36m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb971607310b4d8597406b877e5c25b2 : FAILURE in 1h 43m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e65f231f411343a993624ff38d454231 : FAILURE in 40m 12s","accounts_in_message":[],"_revision_number":18},{"id":"5c2466c701f0d63a5a2a28d36758609b7c76bdd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-23 14:33:05.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/7d9fe64a48c94986a7f63fc0de3a9a41 : SUCCESS in 40m 06s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"70abcfa34b30a44b3a2d1ea47a1c3016324170a1","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-23 15:18:33.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/204 : SUCCESS in 0h 41m 2s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"d576fa338b548cbb26072d830120e9885b7ba867","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-23 15:22:36.000000000","message":"Patch Set 18:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/565 : SUCCESS in 0h 42m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":18},{"id":"16fa3b8eecf48a2d8a15310547a8b972664a5ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 15:26:02.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/966ff2bf1f004e1e855f7210cf4e3d2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4367236026194e04a5ad8fed030611a0 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e343fa180024166b491789d565f8c8b : FAILURE in 9m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a8c30cedd04d28b0569d569976d80c : FAILURE in 9m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1b8a858ef22b4816ab1ecf68e8b8aa52 : FAILURE in 10m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ba4de6cbf14cf78ea6467c3c050770 : SUCCESS in 14m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/efda40289d814266b9076b7bf18c7a42 : SUCCESS in 35m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62d66e258e144b3e85076c6760973fec : SUCCESS in 32m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7fdae4e4be34a59b74f8cdcf46da845 : SUCCESS in 1h 38m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c95f0b53e0f440929940fabc6514de50 : SUCCESS in 1h 07m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/517e0ceab49c4299ab93595cb064482a : SUCCESS in 5m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f738bedcf9d845f7975c401f0fbab770 : FAILURE in 18m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1efd97c3715144bbb7d64c7df922012f : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2384cc2242540eb81896a584cd71f67 : SUCCESS in 5m 35s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/9b1f94d9c3c34612a94abef42cffabcd : FAILURE in 14m 58s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/17e4dd458ae241f3938da62ab65c0cc9 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/99b9b45c83ed42b78a08e9cb899803c2 : SUCCESS in 6m 42s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/646e9330ab094196b71a743f70ac1714 : SUCCESS in 5m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/07736165467e4e3797d2156c361a780f : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d6f20644a52429aa800e0320ea9c247 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1665e75d683745798c1ed2b718458881 : SUCCESS in 2h 00m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d3a022bd87f34fb5aeda3efe01371279 : FAILURE in 2h 36m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e76b87eb20d64be4bca8327c7f6dbf01 : SUCCESS in 2h 02m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0dfcce468017435ba7d5e997883541fc : SUCCESS in 2h 04m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4616e44109d748c4b96dc34b954fd31b : SUCCESS in 27m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a5e5c4457e844a6d906fcb5e1dff8ea5 : SUCCESS in 1h 12m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/faa0258a5b2149ff9f8234872bcbca7e : SUCCESS in 1h 39m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8cea8b010cea4dee9226c76c0977fde6 : SUCCESS in 43m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d97333b3da16420b8c772947cbea0af5 : SUCCESS in 52m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/148045b90cec48bd84ec1bb699f86017 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":18},{"id":"22ce4b35923910d4fe1431e60321080ee76e5c83","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-23 19:18:18.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/aacf2fe0ed2342de954951412e299f85 : SUCCESS in 57m 45s\n- cinder-storpool-tempest-iscsi-multipath build/a99d58089eb64e928c7f968d07e7178a : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":18},{"id":"58064603f42ca3d7697d7a2a6f158fe87f5679bd","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-24 01:02:10.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"23d2f896c027f8035e9b1137042dbdd245f0a2ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 01:27:26.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acc51ef8df94e3aa69aed7b747a53a6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94daf0ec600e4d6490946e06b86feb2c : FAILURE in 16m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e45b54da3af94fca8bb363d34b9fb77e : FAILURE in 17m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2ad784e037cf4a218acc97c9e1ee66f7 : FAILURE in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"db0ad177370690ab6431c1e3fcdf580b2dde3305","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-24 02:01:17.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4496/.","accounts_in_message":[],"_revision_number":18},{"id":"e993283d0b4ac33524440f324542f7d599625512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 03:29:49.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d87b4ddf32e64d9a9527d28f92239e64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c8b9cac5ce429890f82b53b5e6aa91 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ebcb3fce80b47a1b633751350a4a286 : FAILURE in 11m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bbaf7fa7b024eab958d9a410887fa60 : FAILURE in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1893a3ab70a240e895162ba108a36501 : FAILURE in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea09afea55d4dbd8a1f4fbb9d0c7c8f : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/6370df973ed349df9bf2881b67dc1338 : SUCCESS in 1h 05m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f5af75cb01646e8bf6c87444b42793e : SUCCESS in 1h 05m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3f7bd174d55488ba79f7065b0eb40ea : SUCCESS in 2h 05m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ad70e1eabb8415495442f340d0fab8d : SUCCESS in 1h 05m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6632c835f2cd4096bcd8a5a125c520a8 : SUCCESS in 6m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0a234dcdbe764d398fa2054d0d7156b7 : FAILURE in 19m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04039a3cadb7478db62864c17587ec92 : SUCCESS in 4m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/759109e3767d424bbb82c53fc097a61c : SUCCESS in 6m 12s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f241da42b0e3436480c124585dd4cea7 : FAILURE in 16m 55s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf70d54ca934466aa33ce61e0cc188d8 : SUCCESS in 9m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c1c6b80a41374292bb671889f54d1329 : SUCCESS in 9m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9b5c41a19f364893aeeb0b808d7eefc9 : SUCCESS in 7m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bc53403d8af240d396e579faa44d44f5 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/356381288cb34f3c80f8e0ebe8e7e64e : SUCCESS in 3m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/201f49a71ddb4a95864ca575dc2fa881 : SUCCESS in 1h 27m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/90d0219585f94ef68df4c56e06f061ed : SUCCESS in 2h 20m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e4f56a0a009f4ec3a6aef47d70385e68 : SUCCESS in 2h 11m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/00d1e4aef53e45f280f5710a1d089c52 : FAILURE in 2h 21m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/549fb97b51fc4b4f97e03740d1060eac : SUCCESS in 1h 47m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24ab9fb37b91470b9152606b8b2da144 : SUCCESS in 42m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79a76aefe8944987bb8ad5e89f25997b : SUCCESS in 39m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/72410d2fb20a4844b025fe8d842ca68a : SUCCESS in 1h 08m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d72361d2fea04cb29658bdbeb07941c6 : SUCCESS in 33m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49f37a752ece49318819e04418b6f76e : SUCCESS in 47m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/203dd8a8b93f446cb1c9a4cfdb927be4 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":18},{"id":"dbebdd955681f2c9181966b4eb283b2ecee51408","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-24 03:48:56.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"1c222b3de648865fa01cd5599f8ab1b4839017ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-24 08:27:22.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/81ae81e55b0c4cca9831c275dabd55d4 : SUCCESS in 51m 36s\n- cinder-storpool-tempest-iscsi-multipath build/d0566fc749ab4417aad3233729a20cfa : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":18},{"id":"19c9b0a9f254ec4383158dfced2a8dfc2f53fbca","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-24 11:32:56.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-18-fc : SUCCESS in 1h 13m 11s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-18-iscsi : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":18},{"id":"290cda51f574fc8e3c685f1a06a3df80dc4d12ab","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-29 14:42:58.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/977851/18/20260629174417/NEC-ISCSI : SUCCESS in 5h 33m 17s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/51/977851/18/20260629174417/NEC-FC : SUCCESS in 5h 58m 38s","accounts_in_message":[],"_revision_number":18},{"id":"ab7caf187a13fd45d2d2a3952fff311b882f3215","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-09 14:57:23.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"e8bd402fff422dc63d6a858875b51971b83c9017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-09 14:59:37.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"0138df8560417b791e95a0e1063c84de3e3bcc2c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 15:05:12.000000000","message":"Patch Set 19:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/830 : FAILURE in 0h 0m 11s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":19},{"id":"8cbd8139f07aeb6bd957679ed6b5bfc3d2258647","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 15:05:19.000000000","message":"Patch Set 19:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/474 : FAILURE in 0h 0m 18s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":19},{"id":"dbf1d3731df6925aeba81450a27f1d30d899fc50","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-09 15:45:37.000000000","message":"Patch Set 19:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0cb66430e72e46dfbdf8dfe6d6a22795/","accounts_in_message":[],"_revision_number":19},{"id":"32fcc81ca55585ad9bfdfc322d705b35e78492e4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 15:48:35.000000000","message":"Patch Set 19:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/560 : SUCCESS in 0h 41m 11s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":19},{"id":"884cc41371e9e96675bc7eac35487bd035867b20","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-09 15:54:55.000000000","message":"Patch Set 19:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0cb66430e72e46dfbdf8dfe6d6a22795\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01338506103e4e84850e306b97fe1236 : SUCCESS in 51m 05s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a995604c981e4e759b8012d921af1980 : SUCCESS in 45m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4eabc7f6bd5d4f27b78f895a67cc7f9b : SUCCESS in 42m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d45330904d4f4e068784468ab5c7622e : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":19},{"id":"651331c8c82edc909f9b4171b673180191023135","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-09 16:03:50.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":19},{"id":"d57bcf1ef13679cf3ad85ddb0da5b312695361fe","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-09 16:06:05.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/39a8712e73fe46718dbe885097adc7dd : SUCCESS in 51m 56s\n- cinder-storpool-tempest-iscsi-multipath build/401c621561754a9a8d986ec5536afb81 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":19},{"id":"9a945913842c02d0fca881b90d61316a35ce0b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 17:52:29.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0b46ffb525449aeaa463890dee2548e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441b6b2bed3b4601beb8c6b4a3c1bdf7 : SUCCESS in 8m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2768649cd0a7470789a34d5d7347c382 : FAILURE in 12m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e120726b22e34baab6be344edd8035da : FAILURE in 12m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/671ae9014fe44688a30abd9978e6274c : FAILURE in 14m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4e751f5141406591982d40ac857bb0 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/dfc7c27f3c0e47679983f88042180997 : SUCCESS in 1h 10m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f88c2a861dc648cf9132895552b79ec5 : SUCCESS in 1h 06m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea836574cd184fc3926f70250de6382b : SUCCESS in 1h 44m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3ef07abb33147938009c23f30d77674 : SUCCESS in 1h 07m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beeed3dd0ecd40a7aa6f6837fc705b57 : SUCCESS in 10m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/84cdb0e0607b4cecb1efa342127c303a : FAILURE in 22m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca5ecab0cfa44d9da1e547c640cca82d : SUCCESS in 6m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5b38e3e3f0c44f24ad581512744e3bc1 : SUCCESS in 6m 57s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/b77a86d9efd5429788f6f73b7a0d57d1 : FAILURE in 14m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/038abf2473c743299901a6d055f2f47b : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5f0e71ac280e42cfb51849d79e3fc184 : SUCCESS in 10m 09s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/116120a59fab4b4eaed56b0334108543 : SUCCESS in 9m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19d3d24248b4469a9bf74be22eb0ea34 : SUCCESS in 41m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f264f59a4e2047dda18f09f618021de8 : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ef01b8ba959c432b93697e085c120a02 : SUCCESS in 1h 51m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dcdf7617b1be4ff8b43b1a77a80650f0 : FAILURE in 2h 26m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6ba7f0ed8344900b1fb528bf128ecc5 : SUCCESS in 1h 55m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f444b877405046e0accab5c49ca31fef : FAILURE in 2h 43m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/af9c819d9dbc4022b21e757d20a1d5e3 : SUCCESS in 1h 52m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ab252077ba0d4b7886db6f084ee6e0b4 : SUCCESS in 41m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0645175f5bf644b6971a6891b59fdb50 : SUCCESS in 1h 20m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7bad97d95111442a9e79a144829142d5 : SUCCESS in 2h 08m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/96a8c3ffdf4542e099c9c70a0f273a56 : SUCCESS in 56m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fbfcb02ed154a05b8d213d2d16c3e80 : SUCCESS in 1h 47m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97725cb35ab04bc78ad098143be2c5ed : SUCCESS in 1h 00m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":19},{"id":"ecf23012c257b81bf3af9c9e856d5e6397294d90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:50:34.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94ad762292844d97a7d0fe2f0964d24a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc1726131c7b408fa85c9f19b414fa5e : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea2448a3ce924ee49d0c746cdbe101f8 : FAILURE in 21m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2a549cf3caab4058ae0890351a2a1db0 : FAILURE in 40m 25s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b500ed53939e542ea9dc20516b2e829c28e6751b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 01:11:54.000000000","message":"Patch Set 15:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4242/.","accounts_in_message":[],"_revision_number":15},{"id":"3ec84753c5912d093d5d92e34daabda2b341bf2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 02:23:18.000000000","message":"Patch Set 19:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4751/.","accounts_in_message":[],"_revision_number":19},{"id":"c575ed2d5323a4e567b6390b4cedf7eff62e1880","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 06:59:35.000000000","message":"Patch Set 16:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4248/.","accounts_in_message":[],"_revision_number":16},{"id":"a296b65e3a043118498deee67dae6c4f08518fb8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-10 07:06:51.000000000","message":"Patch Set 19:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-19/FC/ : FAILURE in 16h 6m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-51-977851-19/iSCSI/ : UNSTABLE in 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":19},{"id":"80b1765a1a556b98e62bce27351a5018a71b15bb","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-10 19:52:20.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-19-fc : SUCCESS in 1h 17m 18s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-19-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"2dd927f91cb8a4bbb9d009fa43caedf6d5a210af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 17:22:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4296/.","accounts_in_message":[],"_revision_number":17},{"id":"f627d7bd804a76332cfa2439ed60ed38f95effed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 16:12:04.000000000","message":"Patch Set 18:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4320/.","accounts_in_message":[],"_revision_number":18},{"id":"bc49c71d96b3469ee3556b007b737b0550d1e9f3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 20:44:24.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4575/.","accounts_in_message":[],"_revision_number":19},{"id":"ad026490f959f9452435ffc1b2e01b56be6403c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-07-28 19:27:30.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"e96d9578862370a00b7d78cbc7fea3785cdb7b61","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-28 19:41:05.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4942/.","accounts_in_message":[],"_revision_number":20},{"id":"29d3083385d0281174e388ef9bae0dadb22752ee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 19:45:21.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/cdbac9e2391e4a82bbae065bf24dbd62/","accounts_in_message":[],"_revision_number":20},{"id":"b2f9581b9e964fbb7ee186fde77081b5a08f8220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 19:53:25.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e571b03172449e9fcb248198998ba5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/938419a819e2476e8bd0c2408df80d2f : FAILURE in 12m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e04ccc842504bd893cab994d72cfc8a : FAILURE in 12m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/87511766fdc147748cf29ee2abe79d30 : FAILURE in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7b3b821a8d44c4c4a4d527f43047439531773855","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 20:15:41.000000000","message":"Patch Set 20:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cdbac9e2391e4a82bbae065bf24dbd62\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2889d2c93c7a468ea7f02ea0e1c04094 : SUCCESS in 41m 13s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/35f22d6b8d6a44968aa7e8127c3c7a7f : SUCCESS in 32m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/14ca3daaf8fb4c0d877b2763719105ce : SUCCESS in 47m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/555855d1e58a4f95982090bc660f66a3 : SUCCESS in 30m 09s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/51f694b3b8da4f2b96647a1737040fa5 : FAILURE in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"53c12d8c66828d205e7c92d1da9db6bfeeb85c84","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-28 20:16:59.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/737 : FAILURE in 0h 49m 9s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":20},{"id":"403b72c6a6e08d861d84e8a3f4fe4dfcd8694c36","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-28 20:17:12.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/645 : FAILURE in 0h 49m 24s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":20},{"id":"59843b69453d8eda33934818f18215daa4515e08","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-28 20:23:49.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":20},{"id":"366b3bf2bd0e530fe7735c45ead47602eff43973","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-28 20:36:59.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1003 : FAILURE in 0h 55m 30s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":20},{"id":"6f69fcb4db8ed307dcfc7cb3a2decf6c7c9c4094","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-28 20:49:17.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- cinder-storpool-tempest build/35130eeae1714ad5be47609a7b021bd3 : FAILURE in 46m 11s\n- cinder-storpool-tempest-iscsi-multipath build/9ddfa98780654ce0b0a163ed2fb3e0af : FAILURE in 53m 31s","accounts_in_message":[],"_revision_number":20},{"id":"94b05cfbabbea047d98cd1a59d84f3f7976272d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 21:46:45.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff266669ef04e7298f0e8f878908271\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26fdba7675f047d7976a9e2e635546df : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2433cc1260d04d0eb357a1f081970fc8 : FAILURE in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b07e1f6d3414604ba2ba1122037db9c : FAILURE in 7m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6f983f88326642e7b4a959145e52f001 : TIMED_OUT in 41m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d000cf8a190c41c78ab53a2f2ad454d7 : SUCCESS in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/cbc5195ffb7f450195075f84f2693ed4 : SUCCESS in 1h 03m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69e35ca65a7542978ed01ba7b4729a8b : SUCCESS in 30m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0d87ddbeacf4fbd9535ff5d954bf0ec : SUCCESS in 1h 26m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c237247abef4a019902e2c73aa142d9 : SUCCESS in 56m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c1b6f24f95442368b17222530214970 : SUCCESS in 5m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e3243d08a4a448090e1e8c1363d38ad : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/668b623c0b6745b1a9b431aa02b32c66 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04ce5c52352046a68e1f680c477c3a01 : SUCCESS in 6m 43s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/02d2342af9af46cb9950be079420097f : FAILURE in 10m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d307777b64464af0974c49c3a39a84be : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/52535407f77a499fba3920f5ad1bdfd9 : SUCCESS in 4m 38s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/59dab0608c22470a9c1ae8c954a930b2 : SUCCESS in 9m 31s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dd1831393a7240ec949930a28772f238 : FAILURE in 40m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55b36570e9ea48cc93c07a04da0c1430 : SUCCESS in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86388ccbf5cd45f5b261d06925aa53ab : SUCCESS in 46m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3ef99e5d98b94a82b9ba01b55c94dae9 : FAILURE in 2h 00m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7c8d4dec15945e5b011ed1e03752188 : SUCCESS in 1h 36m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bc96ea4cf73640a4b20e995a49e55978 : FAILURE in 2h 03m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fee9a14736364b9eaf3f98622015a869 : FAILURE in 1h 30m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6600d70f531c4055a59baa149ad7f17e : SUCCESS in 20m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f4dfafc4005b49d0a68db69b59d8be6d : SUCCESS in 1h 16m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86572b01c9634c2cbe08185c54beb5d6 : SUCCESS in 1h 11m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca0d98d4460042e59a0f944c959dc02e : SUCCESS in 42m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0c76ccc2a9ea43aa88e72d918e622e05 : SUCCESS in 1h 38m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3702f388fbe34bf68614b15815d36555 : SUCCESS in 30m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"563d7f40474b5acac3dc1c5b81187532c0e8de21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-05 12:17:20.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"52821f88dbc3768413d0d8ac66ca28897328eb0b","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 12:20:42.000000000","message":"Patch Set 21:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/824 : FAILURE in 0h 2m 58s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":21},{"id":"c7f86f335a93697594e506f53b7752a02100daa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-05 12:22:24.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"92dfb3225b081f9840f955360b59acf66df18813","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-05 12:22:41.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"cc89ff86e7f259b54b491c00293fe465771e3ac0","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 12:22:56.000000000","message":"Patch Set 22:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/825 : FAILURE in 0h 0m 10s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":22},{"id":"de842995479a056819ae67dbdd85e16b86940296","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 12:38:39.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc9559b0ebb4b0d9309f1492b160944\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a42a4e4b5dd4f85950ebb2705fdc437 : FAILURE in 12m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cbec2a7733d040998bedf9d5d4004bb8 : FAILURE in 14m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7f7f513a5b1d4edf9c9ca823a69d902e : FAILURE in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"3ba5e42070ba70925b797c413bb1e34f2bd2e3d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 12:42:22.000000000","message":"Patch Set 22:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5036/.","accounts_in_message":[],"_revision_number":22},{"id":"ed5b0ef9a43bd6b9db7ae7e68bb62cf51a1f273c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-05 12:43:34.000000000","message":"Patch Set 22:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4860/.","accounts_in_message":[],"_revision_number":22},{"id":"ed5f6bcd100601f8e40493f3314fc9a17986fb9e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-05 13:09:39.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ef23aa65c18f49fb9f0d3ba815c3b9fb : RETRY_LIMIT in 25m 35s\n- cinder-storpool-tempest-iscsi-multipath build/6b1b394df73c4c5fa5e057560bd050c6 : RETRY_LIMIT in 21m 15s","accounts_in_message":[],"_revision_number":22},{"id":"6628c1d65b8d48a3d97f0277b2aa3ceec47ce323","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-05 13:12:46.000000000","message":"Patch Set 22:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/732 : SUCCESS in 0h 46m 59s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":22},{"id":"9e01121d17ae2f3c36519b2351caeae2080b36d3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-05 13:13:40.000000000","message":"Patch Set 22:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1093 : SUCCESS in 0h 48m 22s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":22},{"id":"a97297a6c4be798188f9f6a8a0ff4a13d383dd85","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 13:14:26.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":21},{"id":"fca79ad7d59a0f894e000f6493d8539b02650dd6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 13:18:04.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":22},{"id":"3f7ea868371a00964789a4ff60f0d6d71b4fea46","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-05 13:18:59.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"52d2a4029aaea0ac4e9d4d9fe8244774df91f9b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 13:35:33.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b1e1eb7ac2b741bd9c5060513faa86cb/","accounts_in_message":[],"_revision_number":22},{"id":"45c034c8b5d43e16a701428c4e3be2edc2ceb8c3","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-05 13:36:41.000000000","message":"Patch Set 22:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/828 : SUCCESS in 0h 46m 2s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":22},{"id":"dac29a9d079900b4ee842e82f1ceccbf7f9835bd","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-05 14:07:34.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"cb68d230e02cd13132963518ae867200735d992c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-05 14:09:06.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b1e1eb7ac2b741bd9c5060513faa86cb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/39bd090660d24b7ca73050d693f0e15f : SUCCESS in 53m 46s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d88ec2e89044df3a2ba0cd17e678450 : SUCCESS in 41m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7409a9fc7a3a41d092109293d73a3caa : SUCCESS in 1h 03m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81f68103eb1448e8902d5e3deedfecd4 : SUCCESS in 46m 09s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/59c6c3267d154090a0af56fc3a3712a9 : FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"03de402263834f5b2b2d4a21cba2ae684e7293f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:11:26.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eea45bd9a1314fdcad18075d31c7da87\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38dd063d04534e62b319eafad06433a9 : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d264a8b151f640dbb805cc70c50bfde3 : FAILURE in 10m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7781d90785f40838475f80975ba3a19 : FAILURE in 10m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6558283daaf499a91b9a975794c7205 : FAILURE in 13m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71742cdc28cf44c08670a5be9a63bb2d : SUCCESS in 17m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/99b349de919d4c3d933e5db07f64e2b5 : SUCCESS in 1h 01m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecf2707326934842ad5e0061a9571c68 : SUCCESS in 1h 17m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/947bff68499c483ea288e982999bf1b2 : SUCCESS in 1h 44m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0030cc657d04a79b933947eaff5c695 : SUCCESS in 1h 01m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c911844f2a8f4be2af108d922691f83c : SUCCESS in 5m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3996b81594ff4504b1e15dfe20e8ab7e : FAILURE in 15m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e77ef5162e44d2bb004ac94d0c12012 : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08e15552b5284f5c987f6b895c2ebaed : SUCCESS in 3m 15s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/0fd54d4bfe764fd19737ac0b4267afc2 : FAILURE in 9m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/007386e98579471092ddb2392db0923a : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e3d3a1bf0bcf44ac951baf09a293e8ee : SUCCESS in 5m 55s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d224b40adaea47cf9c6bed36a933e493 : SUCCESS in 9m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f2e505ca7295451086fd0c4949043e95 : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17812a86c9ff47c08ac66f277f72225f : SUCCESS in 6m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/81b8bfc01d104d3b90eae7664d6bf6b4 : SUCCESS in 1h 21m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2afc6f47b6d14844abbfa6082ecb7c73 : FAILURE in 2h 18m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ff44fc46e53b4522b17c34ac565217d5 : SUCCESS in 2h 15m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e995f92b15424c6ca4132df869097f96 : FAILURE in 2h 46m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0b3b3640c2c44e2e8c209a4835d465f3 : FAILURE in 1h 36m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/385da9a43ea64710b8a73578f8af1383 : SUCCESS in 22m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b626a9ea60de4a18b0e3788edfebf3f3 : SUCCESS in 38m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/016c1be13b254c85bb61486394dd0906 : SUCCESS in 1h 18m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a95727622bf84d9a831a58cfa2d88221 : SUCCESS in 50m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bb65e66a9e124778b2da725dfab66dd1 : SUCCESS in 1h 53m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80be0760c3034e8d9199a7c13e3fde61 : SUCCESS in 29m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"af71ab5c36530119643d8b64f9267363c626ce5d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-05 22:53:05.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f53bee3c69cd0376548e24b9472fba4ea35053e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 04:48:51.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"dea8080255c05107fa43b2345f4a3376b131cbb6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-06 04:51:44.000000000","message":"Patch Set 23:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":23},{"id":"ab0ee26f1609a2ba494e0a9ec13b814eee616a04","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 04:57:28.000000000","message":"Patch Set 23:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5050/.","accounts_in_message":[],"_revision_number":23},{"id":"573a222a2f1b24bc88052259001282ff6b594e76","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 04:58:52.000000000","message":"Patch Set 23:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3f762a9cdb0043438fda634d53cc1a6c/","accounts_in_message":[],"_revision_number":23},{"id":"6f3e3efc79de138ec66769d4ade696730b36077c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 05:02:17.000000000","message":"Patch Set 23:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4874/.","accounts_in_message":[],"_revision_number":23},{"id":"112dae832eb907c4c3004be57aa2850163bc3404","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 05:11:23.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e4551b15534cc298b94b615f2d0bfb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f34b70c9f654039a35c2d3328df389c : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f8e7451386249669961911d787e5fe4 : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c676acbac9164980be63584d1926920f : SUCCESS in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bd7358373961803112427479f0c64ed6c645e41a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-06 05:39:37.000000000","message":"Patch Set 23:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1107 : SUCCESS in 0h 50m 26s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":23},{"id":"8650d1125114da8660d65495aeba63fd24bfcf2a","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-06 05:39:38.000000000","message":"Patch Set 23:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/746 : SUCCESS in 0h 50m 26s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":23},{"id":"7357a8ee6e35d57b9c2edd3688d5187ec574b5f8","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-06 05:39:40.000000000","message":"Patch Set 23:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/840 : SUCCESS in 0h 50m 26s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":23},{"id":"802341fe8cb819c15045909eacb93ca0067ecd2b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 06:12:57.000000000","message":"Patch Set 23:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3f762a9cdb0043438fda634d53cc1a6c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6aafa2e9b11741a7883db20de3dd7cf3 : SUCCESS in 39m 49s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77b7d1cfbe8a4552a8f801b7ec86a8ae : SUCCESS in 35m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/327426abe76b42f28ab1d8db50074c01 : SUCCESS in 46m 17s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c8e7c3aff71f4726b9c3405613da24f0 : FAILURE in 6m 21s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f5b540f384e4ab490e55f7ff7736084 : FAILURE in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"4922f214b5c515b9d484f0f3f74514b0ed7c343f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 07:28:02.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfd984d7ef3c492ea668b4f4c8241217\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d856438d40ff4c25a1be56196c220613 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1ce20f5c6094ce59cea901427c12a3b : SUCCESS in 9m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37f35ad31cc94abd98d2d0f8db7c2e67 : SUCCESS in 10m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8e43240987c478aa7aa1522dd1b6e3e : TIMED_OUT in 41m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e81392b39ec405bb22072592dc5d15b : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/da2270e398b84ec88a58487b47d932f5 : FAILURE in 29m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab332b53bf01496f9c29b154af94deda : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fbbef85854c4d28891be9268294ba6c : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cb5c5286569416a9570fb305b49b4fe : FAILURE in 1h 01m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a0ed4d658664d7297d0247bf57698b4 : SUCCESS in 6m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f711306e7794496eb77524a6d2e86cb7 : SUCCESS in 19m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b6afda6b42804e3f949cc60b7f937644 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3265dff46654e5e8442dc8bf7b29b1b : SUCCESS in 6m 20s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/700586aa639241ee8a0d14cc6f4c8ea7 : FAILURE in 10m 38s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ccc54c68c2e94873b614a5b18e4fe616 : SUCCESS in 7m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b1f703e1146b4f57902c2efb6c0f5859 : SUCCESS in 4m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1ecec480ca204582a4fb1180a03138ab : SUCCESS in 8m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/362b1f8e0bbc42be8db92765f857ee6a : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a4669b052e2c44e9b91931bfe6f5577c : SUCCESS in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d230fd970134d2497a7178dc4ebb601 : FAILURE in 29m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/955159a5ebc44ac3a36838427f3cefcc : FAILURE in 1h 44m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1a6b20cec37484cb19ad415c96a7820 : SUCCESS in 1h 28m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/efbaba01e75545b4932e81fe3255c2e1 : FAILURE in 2h 32m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/824309aeb3174479b48d053a914b45c0 : FAILURE in 2h 03m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a0bb13bfcf3c4a84a59e66a5927b52e8 : SUCCESS in 17m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db041fd1876e4064bf3d23e126bd8032 : SUCCESS in 41m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8158f40f62cc4644ae58a7f007035aaa : FAILURE in 13m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/07aef966e26a4f17b76bb63554fa808e : SUCCESS in 26m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/25158d60252f405db24f15d72455cec1 : SUCCESS in 37m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a26890c3e1364c0bb9a2580d3688930c : SUCCESS in 39m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":23},{"id":"129f77e89716faf9e66d63845e70b46fc47a177d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 07:51:07.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5314e29395124e15b4d02b8b1d13eb68 : SUCCESS in 57m 20s\n- cinder-storpool-tempest-iscsi-multipath build/0e632a5682d54756984fd35946ee06dc : SUCCESS in 1h 01m 16s","accounts_in_message":[],"_revision_number":23},{"id":"101bc41880e9ed4982763d1e39e589ac61a6bd09","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 16:04:52.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"e7ffab8c8fcf42c41b832d43e75837769c7e6d80","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 16:07:29.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"57a6364057db8945a4f02a6037579271dff81414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 16:33:11.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"6a8be782288e02623e006fc6fee648095cafa916","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-06 16:36:04.000000000","message":"Patch Set 24:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":24},{"id":"53ae5afd305e1b2dc3e71d4ad789f55c75884fbf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 16:44:36.000000000","message":"Patch Set 24:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4878/.","accounts_in_message":[],"_revision_number":24},{"id":"70f2b89200e552d9157f31b23997c4470673c124","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-06 16:46:59.000000000","message":"Patch Set 24:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5054/.","accounts_in_message":[],"_revision_number":24},{"id":"f4ccafe11e69e0d44e6c069a4fcfed62eb49307c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 17:12:34.000000000","message":"Patch Set 24:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ae6b8a72011f4117a8fa1792bb7427e6/","accounts_in_message":[],"_revision_number":24},{"id":"074e7f57c79db5a7aaff83965d343664f55bc5e3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-06 17:22:53.000000000","message":"Patch Set 24:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1111 : SUCCESS in 0h 49m 21s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":24},{"id":"9dace94b82bb3cb20a628e26ce89b4b58f24ed7b","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-06 17:23:35.000000000","message":"Patch Set 24:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/844 : SUCCESS in 0h 50m 5s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":24},{"id":"bfbdd7828df7a890d907f8c4e3f425fd00f86854","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-06 17:23:40.000000000","message":"Patch Set 24:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/750 : SUCCESS in 0h 50m 11s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":24},{"id":"db52102a675409382b39b4943d31e6a7481cbabc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-06 17:45:21.000000000","message":"Patch Set 24:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ae6b8a72011f4117a8fa1792bb7427e6\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76c9508bf9ac4929bca32fe0eaaf8676 : SUCCESS in 51m 14s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d58af3e3ce94868b330b2e87ca802d8 : SUCCESS in 38m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0bd23a6b4fe4afeb2dd717f8b747fc7 : SUCCESS in 49m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac139eb31a424c8e957b31742b034551 : SUCCESS in 39m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea2b207a4eb745ddaedc42ddc543dc15 : FAILURE in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"b251a6e1f44a30ac1cf7b03fa9a5c578d43ae129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 18:03:37.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bada6af18c2405b9ecf1ccaa683ebcd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/734c042b35724a8e9692f51f5829c491 : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b9bd8fd49b074a10a106eeb7e21c46e5 : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df443e2ed77b4f7c9a105d67b7752e36 : SUCCESS in 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"2bac7d5cd454d64b6cee304525a0e8f9c737b01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 19:11:08.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0578438276e480ca338a668292b3858\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108d2232644442cbb770c1907b37a940 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1a779a4f1d74200ae7270b3c74a57af : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dff948d2559540139c3e7b66e8d1d00e : SUCCESS in 9m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c7e3de55dd643889b7ed7f8792683e3 : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d927767b60064440b5e3863fbb2a1e65 : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/0e83de086fd646799370b0fd2c621515 : SUCCESS in 1h 09m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec5ed0cf14aa4e52818bb0d38f8cc443 : SUCCESS in 1h 00m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54194a5f491a4dc7b954a93dfeaf27fc : SUCCESS in 1h 25m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b3ec4a660304a7db13d8e3194593e89 : FAILURE in 37m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba9b98d45bd84272b8f8b60cbaf43e8d : SUCCESS in 9m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3fc83e9d77584cfa91ded327eeab48cc : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4663531237854573b318a401506b3190 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/50f40b24eeec4153b40b6083316dfd0a : SUCCESS in 4m 43s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/0a1aae246255448bb5fa50a23501cb87 : FAILURE in 13m 38s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9ef46d5c101e4addbe4d4b4e986b54fa : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6cf3a5d251c947eaa935bf43a371ad02 : SUCCESS in 5m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/fa0a29f6624c49bf97aff1512266fd31 : SUCCESS in 9m 15s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/772a60e2858944e5b5d01142820b1f54 : SUCCESS in 37m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7baac5970f5a46629eda93c2f8c8b7bf : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d54d8ee81063459eb9c03e22a87c658d : SUCCESS in 1h 38m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8ef102efb72343d6a2e8639883ac01d5 : FAILURE in 2h 12m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5cd11462c80b4cd486ecc6a20a95b879 : SUCCESS in 1h 46m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ba598d5ca165457a811826c2c9451ce0 : FAILURE in 2h 30m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0f1c065565ce4e85be9a14af3a339bdf : FAILURE in 2h 08m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/57c19e48aa884e2ab4907befc4520a7f : SUCCESS in 27m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3093c1b7ab06408b92232fea21cc1a2f : SUCCESS in 40m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/432508aaf41f47398f01dda6f7ecb1d0 : SUCCESS in 1h 12m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c562b655f3144fe6b29a5f2ab5cbd54b : SUCCESS in 35m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37aadecb3aba461ea93bbb6a171226d8 : SUCCESS in 1h 29m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/778ba7a104074a279f6ded827618d7cc : SUCCESS in 36m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"58eb57f22d40a03e1e185f2ee81e03190ee4ac76","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 19:48:02.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"b1664dff89f3b118567273c69f27ea367542c35a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 20:27:03.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a4a869b4a98458d84505dce7226d9dd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18537708144041579f81938b2c230689 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55e50a7d20124f5690936a29b8fdcb93 : SUCCESS in 15m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c79bec75e49b4241b1a53cac6c0907c9 : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"83ebe57ff0ec7d667d636d47007633ec78f37035","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-06 20:31:43.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/dc194c3556b44f638ba6f166d47426d1 : SUCCESS in 46m 03s\n- cinder-storpool-tempest-iscsi-multipath build/2322eded71974271a790dfc7e19a5fb5 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":24},{"id":"716ad99825c2f7f750201ba237f53a313c2b9677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 22:04:51.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63cdee721ab74602ab219efb3482f2a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2f03464c4044c8b4ed8551b568f327 : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ed0fa6aad8b42b19843ff7f5de006f4 : SUCCESS in 10m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5325fc6312ba48cfbd33483ca347a6bf : SUCCESS in 11m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9c6f832f5d44b688841832dedf0a140 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0e853c429040e4bad95b582ed3ef67 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0dfbefa6fbd54b098573458a7c7b9114 : SUCCESS in 1h 05m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74bd46a9ef4646f3be8a7f40dcd52f8e : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2f3ca79b37f46538fe1ca7a278e5c4e : SUCCESS in 1h 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d10a78bc4ff4321913f6da3919570ed : SUCCESS in 28m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11a9a750b1304b3a8d6f1424700daf8b : SUCCESS in 8m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ae573cf1c4e4ee294dcc57549043313 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4e00f669a133419baf96433a2e8a23a6 : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9cc4c51f761f4a1e8c16ca1bb4d6cd11 : SUCCESS in 8m 33s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/62cc1aa263f0453397c503cec308e61d : FAILURE in 9m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a5f78c9f7b144116a32c4ee1c87f148a : SUCCESS in 7m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/867b3e4977eb47388f0df5e1ac6844ba : SUCCESS in 7m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0bc32f550bd649f6a4173b3078f09ae9 : SUCCESS in 7m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3924cb7737e84731b9c6ace4475ab6db : SUCCESS in 38m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a13e986367447eda81f45cb5647ea6c : SUCCESS in 4m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35c4cf5fb9484a198f05e46a58c41e10 : SUCCESS in 1h 37m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05a7c033259540d98fb30093eccf1415 : FAILURE in 1h 59m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3272bdb155094d5e913e693eafe31d16 : SUCCESS in 1h 45m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/323a820f43e547dd83a1c4b27b299fd1 : FAILURE in 1h 59m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a1c8f2301ccd4cf6a6c920876ab10664 : FAILURE in 1h 57m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7487394db01d41509e14ec3565328086 : SUCCESS in 26m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e74dad0831a409f989d77c7133dcf27 : SUCCESS in 1h 16m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b59c542ce0c4ea9b7b2641aa37534c5 : SUCCESS in 1h 21m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b2a0cdfe3f7e4ae09d400f2500366743 : SUCCESS in 24m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2efcd3d166924a4f8252ea7886e8fbb9 : SUCCESS in 38m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05510ef4647b47e99dcbe92504d393a7 : SUCCESS in 55m 24s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":24},{"id":"21212a3c36850308045350701b052243fcceb90e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-10 20:49:40.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"d09022b1f5e16e790879f407c6f78511f51bf4c0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-10 21:00:10.000000000","message":"Patch Set 25:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4910/.","accounts_in_message":[],"_revision_number":25},{"id":"90d816b4dcd87088d96fa7c030622ca45f35db28","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-10 21:00:28.000000000","message":"Patch Set 25:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5086/.","accounts_in_message":[],"_revision_number":25},{"id":"1b0f9144cd9f779c8b016b92d4b3a0b003921704","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-10 21:08:02.000000000","message":"Patch Set 25:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/7b45092bb31742babc97321497f8bfdf/","accounts_in_message":[],"_revision_number":25},{"id":"0bd8a5b19b64b182d0ebe2cb7b797dba8a34cddb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-10 21:15:22.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- cinder-storpool-tempest build/36fd89cbf8b24392a8562661fe578a4f : RETRY_LIMIT in 24m 06s\n- cinder-storpool-tempest-iscsi-multipath build/4929addb2b09443584e028f45f2c3fff : RETRY_LIMIT in 24m 08s","accounts_in_message":[],"_revision_number":25},{"id":"194797f1b3660857fee79a1580a25ae50ddcdac6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 21:20:31.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ae7c86ea7344e78bee3791bbf33452\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7b108a8c7214cec8d892f4fecafb243 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78f2c5e0877a41d08508a8d506fe049e : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac7c62f1553d422b8c62b6687c0a1412 : SUCCESS in 28m 41s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"56b65190026e43d431c7b4b079603edfaa0239c3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-10 21:37:23.000000000","message":"Patch Set 25:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7b45092bb31742babc97321497f8bfdf\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/876cadba93464d2d94af29cc95551a30 : SUCCESS in 40m 16s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f383082656d0428683549cfec14cca28 : SUCCESS in 34m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d2b0dc7d6404ab58437564c80ac3cdd : SUCCESS in 46m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac6c89d020c74466bc1f823df897ab36 : SUCCESS in 31m 45s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/044b19f4daa64c17ab106eacea744d7f : FAILURE in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c320083aa3e5215d60566ebc2e18832bb2212fba","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-10 21:44:03.000000000","message":"Patch Set 25:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/782 : SUCCESS in 0h 49m 28s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/977851","accounts_in_message":[],"_revision_number":25},{"id":"780d8eb54ed3581dd0b45884d7f37e8fb55774cb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-10 21:44:25.000000000","message":"Patch Set 25:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1143 : SUCCESS in 0h 49m 33s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/977851","accounts_in_message":[],"_revision_number":25},{"id":"79da5848b1c0d826fa2deb0a5b8b24256cadfe8c","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-10 21:44:54.000000000","message":"Patch Set 25:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/876 : SUCCESS in 0h 49m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/977851","accounts_in_message":[],"_revision_number":25},{"id":"b5ebae6f7943b63f2c38c360f9122fd843ade09b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-10 21:46:16.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6320b495e7da077fb7204b50c8d06d736a36e946 .","accounts_in_message":[],"_revision_number":25},{"id":"4f356722b51972d8fbdf17ae383c3b0b0e30fad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 23:01:46.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ba82b3c3904cbeb842aaeac6e31454\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cba1f6976894edc8f809c0671c20250 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5efd49b4cfe3467ca6b2383cc0afc2b7 : SUCCESS in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c872ec0971a40b798d893adb250c97e : SUCCESS in 11m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4965925b6a74aa482971160d64029af : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e782c4e17f4653a65adbe687a03a87 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/25b947c4f57e4e4db1a2f4e0485ba7cf : SUCCESS in 50m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67d1d46f59ce425db3e756316aefccce : SUCCESS in 1h 07m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7247d566e7e6437aaeca7f6bddc66212 : SUCCESS in 1h 25m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5aa0dcd44d414d8f84bb51a4f044fea6 : SUCCESS in 1h 02m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cd6269675ba4b1e845789c94189e907 : SUCCESS in 9m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/12d02685b565485fbafb49c74da7220a : SUCCESS in 25m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/515c3fc443504e5784d02f15f606eb6b : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a78b067afe8a4d3996d933b977f89086 : SUCCESS in 7m 38s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/29d88119ddf241ada5bf8e6f5e39f712 : FAILURE in 10m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a17b246ba1548a4817694872c2488bb : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/17659fa14fad4ca68cd88c868eab95a0 : SUCCESS in 6m 30s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/250f3681313146b898b0b24417a8b75d : SUCCESS in 8m 38s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c8b86473151437ca71fb37cc150694d : SUCCESS in 39m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2aaa76c37c7453aa2337fd88963e36f : SUCCESS in 7m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/711ae43bf36d42979e33d7f96afd38e5 : SUCCESS in 1h 36m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7dfb2e8568064e5c8bb4e05e4ad4aa90 : FAILURE in 1h 59m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/12727e8b48594df298a8757b5a7c3054 : SUCCESS in 1h 45m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ae6f3aa9a4bb44129c01d9dfa31e6f31 : FAILURE in 2h 05m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a937c57090e84ffd89ae68419d6919af : SUCCESS in 2h 00m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/476a5e5f262e4957bc4d609d31146909 : SUCCESS in 36m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3727d7fefa2e4aeca9ae23477d422909 : SUCCESS in 1h 19m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dfeb8dc0ce1b4ea49e857e28d2fa6326 : SUCCESS in 1h 40m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64db335621ce45769dbf845a816c24b2 : SUCCESS in 46m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae0bec7bafb547d58b695e1d83a9e9cd : SUCCESS in 1h 34m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/440686e7295d47239489d8b30eafc894 : SUCCESS in 58m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":25},{"id":"543ed024a14947173baca39e4941cc0b374875d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-10 23:12:19.000000000","message":"Patch Set 25:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_08_2026_23_05_14\nLeave a comment with","accounts_in_message":[],"_revision_number":25},{"id":"8b39e9023b9453cc4da9fc22ecbe34f335bdcfca","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-11 06:58:36.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-25-fc : FAILURE in 16m 15s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-977851-25-iscsi : SUCCESS in 56m 7s","accounts_in_message":[],"_revision_number":25},{"id":"de227dfce9d107a731a3881428fb1c3a5cf6a990","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-08-11 12:39:55.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"733001eb4b82c7d4b6fd65a075e0ee9954907b19","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-11 13:48:12.000000000","message":"Patch Set 25: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"c630929a8ef37990ce8a09c7f45d78cce9ab5a91","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-11 13:51:32.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"b1e314e53fa35ac95f0954976ec0ab5f71c46f22","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-08-11 14:16:07.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"d32f4acbfa32042c7a01f2f7529aaa9a9ccb1f79","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-11 15:33:38.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ab7a0592d6e13d2bef8a07077560796731b420d9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-11 15:33:43.000000000","message":"Patch Set 25: -Code-Review","accounts_in_message":[],"_revision_number":25},{"id":"a5a9f4733979923a396aa8642bcbc35ffa51e931","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-08-12 12:17:30.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"4863bd7bd328713a2e25ca84507b2ad33a87fb15","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 12:35:55.000000000","message":"Patch Set 25:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/7ed7b4a9c4db42f88573c61a3f37b698/","accounts_in_message":[],"_revision_number":25},{"id":"9fcbd2689c5ead2e0d7b978f45db104e40f1bf02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 12:44:25.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce77883fe3e464dbe006a3fec615160\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8dbf9308b63648cfabad8d7649106e8c : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad3e58b3381f40b68fbf82b238d800a8 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7ab2833990b74961b94abc6214f2d410 : SUCCESS in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"70ab8d09ad5194821db100607a4d9616596963af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 13:03:53.000000000","message":"Patch Set 25:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7ed7b4a9c4db42f88573c61a3f37b698\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93605b88b7c341c7a604cf7f08cccdd3 : SUCCESS in 45m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b2b8635b8b1451ebb8f5f3578c8c69f : SUCCESS in 43m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/12d522e3ddd14d088b05a35a6da815da : SUCCESS in 45m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ae72211a804f41a29aaa5d626a348be2 : SUCCESS in 34m 08s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddf59c7947304e828dc61a4e71edb989 : FAILURE in 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"59bd5f262de986dcfc364a45854555332d779d5a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-12 13:15:52.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7c5958e5decd43c788bc77b66bc5e595 : SUCCESS in 56m 52s\n- cinder-storpool-tempest-iscsi-multipath build/56694d0c850c48ac95d9a2b83dd978fc : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":25},{"id":"d5fd84d851b8ad9558d8e84b58a7496637e87720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:36:01.000000000","message":"Patch Set 25:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35727ef8e29a4e68b8784f7728679c20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4feedbc9efea4920addcdaf1c01f501b : SUCCESS in 7m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f77e8f4181f04c3eac88f5b2e04f0fa2 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb8b69a298ff46b0b218e8690390dd0b : SUCCESS in 10m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15eb6a3f9a034a73bbfdb874fa366b73 : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9323b891fe0541e8b4588c22c64821cf : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/289f72aac5c64eed9c6d93ca11d635b8 : SUCCESS in 49m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38a86cb74435426eafd3f8447e4c3543 : SUCCESS in 1h 00m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6ed619ca8907402aa876b8a34459ad4a : SUCCESS in 1h 24m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3550730bcb04ec78d7bacf8653d1577 : SUCCESS in 1h 05m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f11805b60a4133a459ab772eef75c7 : SUCCESS in 9m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/771394f35eef402fb8c044a682716f93 : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/42c71eef454d4b9c988586ec5185c500 : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/259731cfbef747cc92f543aa1cbb46f9 : SUCCESS in 6m 18s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/6a78756c9dc345fd81120af69c2a057e : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c3efcec9ef5f4bbcb2a3516a127f31a5 : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9abf90c7d98b4d8198ca35578085075d : SUCCESS in 5m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/60b736e905ec458c8e6e3ca81107a71f : SUCCESS in 8m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b6a362faad584abb96e9882f5cb05761 : SUCCESS in 38m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bbb882b8fdf4e7ca3d71300d2778620 : SUCCESS in 5m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce0b81851d1b43a988d1ddc4f92e72c7 : SUCCESS in 1h 36m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/35aeff05f0ed4c9b949bd2290bf79845 : FAILURE in 1h 37m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e426eab2a96c4034b374ee797b9793f4 : SUCCESS in 1h 55m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/efa94454921d4fd78a0b91ac030d284f : FAILURE in 2h 12m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9cbff64350cb4fe98582e7fe93878f7c : SUCCESS in 1h 27m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d2568cc58122408da574a167cb1d461a : SUCCESS in 38m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0fea1e1dbc9d4e3f96d439f3063e34b4 : SUCCESS in 1h 21m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/63a1397d7a474f5c9c61c175b3988920 : SUCCESS in 1h 15m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/906b500099154646941a814305015155 : SUCCESS in 40m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69e3bb0abacf45c48b46e3685ea7bb45 : SUCCESS in 1h 35m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/935bf9cb73074d02a28c0c37bb061da6 : SUCCESS in 1h 02m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":25},{"id":"31606f8e8f9e81e628583fb9bc9f469a9923472c","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-12 15:27:35.000000000","message":"Patch Set 25: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"3ef2147f9bcc2bdd1d3e5070dddd08a6841760c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-12 16:19:53.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"f2804f6700944b3e3bd09db3e52e47eea20f0290","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-12 18:04:44.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"ca7b3c352c6e9296178a962d41bd368dad44dd20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 18:05:18.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"9fd40d5af2f322e4820bd30b10da549783a97750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 20:44:21.000000000","message":"Patch Set 25: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d389afb8dc74de3bd514d255f7a6af4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a061842d1d634e1eac90676d3ebd5349 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7396b34c0fdc4a7ca825c4072fefc357 : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cc89de8453946dbb052676b563b1ad4 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb5c72e9fd74ccbbc05d751ea22633a : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/35d9d24f6c4843dbaa66bd16607633f0 : SUCCESS in 1h 06m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a37f17181c247b1a4557efe4cbfb77a : SUCCESS in 1h 07m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0fbd2f9f251f4f70a33373bfc6e6284c : FAILURE in 1h 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b68bc0333e0848d4bbc2cfce527d7922 : SUCCESS in 45m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dada2449310498ebf22118b656d3a6d : SUCCESS in 9m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2dc5aaf267744020a76e0225e4efdaa6 : SUCCESS in 1h 13m 50s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b255d474283b4856901a199a43129dc3 : SUCCESS in 1h 38m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e43b1fab8d7540e7b4b22d22cc4df911 : FAILURE in 10m 27s","accounts_in_message":[],"_revision_number":25},{"id":"18865ce846c0e3bc432ef80686be1495dc80cdd2","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-13 03:01:52.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"ff63abba14d52ddd1a6000c7a14d5215b7ca3550","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 03:57:08.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/96090ae405694b579092deb380929a5e : SUCCESS in 48m 45s\n- cinder-storpool-tempest-iscsi-multipath build/6b7be8daa2874689ab7c6863a4942921 : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":25},{"id":"2916c5b0570e388c2f301d62e15d1529983f4cfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 05:17:57.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed5bf70580da4e2e95c905886354b355\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba9281c1ff04411684299670a1b0a41a : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1094ff5231af492c87f1041fbce0b250 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f53617616e824ac18426f20d5548248e : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"2390ff37d430a6c766c999e5dc818c251d1e6f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 05:22:42.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54d38705586a4803b6f00af45c4f4883\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0609f124878424cb05b425722c9f4cd : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d36f641ef114486bae8f0d18a54710a6 : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/142e7db7faf54826ae8e8c2168ae72c7 : SUCCESS in 10m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32ad5ecfd6fe4360b1c0b35b9bd9ee51 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256dacfb483d4fad8b3c78ea64cd2943 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/67f57fa149294978bf23a947ad2c2bed : SUCCESS in 50m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9bd954ebb284cddb7d3eecb84d6701f : SUCCESS in 50m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa283314219f4cd48c0fe34d2cd1da75 : SUCCESS in 1h 20m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d5a6b2f5630447abb9a935a1eb53e04 : SUCCESS in 46m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd5fd660fffd4a7994d67a9a51dc6595 : SUCCESS in 10m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d93ead4ad6b49b5bb2bd4e5119c5489 : SUCCESS in 30m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92bcee91003143dfbdd1160fb19289eb : SUCCESS in 3m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6ace69f59a454e5fb1eaddf1c3dd9eee : SUCCESS in 5m 54s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a17f07e844ba4a0396e60d0f96f428a3 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/659648a8b80d474d96eeeb7312608cb3 : SUCCESS in 7m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e2e4e490b573470ba6a6247ab172ba05 : SUCCESS in 7m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/dc4122fe12c240f0adec5c63d6a2efd7 : SUCCESS in 6m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/63b14f31865b474c9edf42e27fe2dda3 : SUCCESS in 35m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4bf7aa1d9f904323a3a475f4ba990317 : SUCCESS in 5m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6896d12c7488490e9063c2eeb6222c62 : SUCCESS in 1h 08m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d1713e9e30554b0c8ad081599e144487 : FAILURE in 2h 08m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a138afa8f7164890b1d59f04ec997fd0 : SUCCESS in 1h 47m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/afa42761e55e478fb56b38c21e07f6ee : FAILURE in 1h 52m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/664238eae2b1430594350eecb8c445df : SUCCESS in 1h 49m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d4a3524e8cec4ea2ba155664953c5d6d : SUCCESS in 21m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4bc9845c06b74521a695d20d0ffb6b26 : SUCCESS in 1h 19m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d467293b961d433c807857f37ca867ed : SUCCESS in 1h 36m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7141b69ae3914690a6a26402540fe1f8 : SUCCESS in 46m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71da0bea284e4d4e9eef55a7cf123f4d : SUCCESS in 59m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bd0b72f452c4a478a6f11799b15bfb1 : SUCCESS in 56m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":25},{"id":"0300bd6b69537a1526cd909e5343514c830fb430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 05:23:13.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"98925752da0e1ef802ba0d81c59c91413becd625","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 05:33:15.000000000","message":"Patch Set 25:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_05_26_15\nLeave a comment with","accounts_in_message":[],"_revision_number":25},{"id":"47a193011a8b342b050a12df163c77a317e74363","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 06:47:44.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07c4cfcbf66a46e0a3c895d02272ae9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b084e4cdf3f54af59ef753bed9188b70 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddcc254d2307430398f86e1e261c10ab : SUCCESS in 8m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bc088631603401a82810370787addf4 : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f040b6231b164f94b471a85422c33e93 : SUCCESS in 15m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/05f0fd25b68f47be88251d2be4f26062 : SUCCESS in 1h 01m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2990db70ee81421991b2c1c29e581259 : SUCCESS in 35m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6288078dbec24a94b26b37f898c8d0eb : SUCCESS in 1h 22m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0b5720461ef46b8ba0e8baa92ccc8c3 : SUCCESS in 58m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20654ff964a34a45abbee171b5cc661b : SUCCESS in 8m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/12676c11a8884009aa6f0a0c3c973b85 : SUCCESS in 52m 54s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7504f68b9ac041b0bb5146c2663799d4 : SUCCESS in 52m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25646234f3ba42168546ae78c27fa4ae : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":25},{"id":"7ccfb195a123fc1bdd362f4fd3f7a74cfb58e81e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 06:47:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"d841e2904140235eed714838ce9c43a5479b00c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 06:48:47.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f2be3f0554a4e318cd871c02c482547\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43e80125e0c4bf3b4cfbe8a5cbe0360 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":25},{"id":"11a0a50a6eea6dea8a5685aab33b32a005466aaa","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-13 11:12:24.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"c07c49c586d1667d3e11a7142689894fd9196c74","revisions":{"c9ad5c6448210c1b90739bbd4b53a073d171bef3":{"kind":"REWORK","_number":1,"created":"2026-02-24 18:39:00.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/1"}}},"commit":{"parents":[{"commit":"b6b50ecbae86642a66feed08ab563d2e79ecbac0","subject":"DNM: Set threading mode to enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6b50ecbae86642a66feed08ab563d2e79ecbac0"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:38:33.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9ad5c6448210c1b90739bbd4b53a073d171bef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9ad5c6448210c1b90739bbd4b53a073d171bef3"}]},"branch":"refs/heads/master"},"50e1083f496bac168ebc6fb1e5e16d6d9573f8db":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-24 18:41:57.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/2"}}},"commit":{"parents":[{"commit":"b6b50ecbae86642a66feed08ab563d2e79ecbac0","subject":"DNM: Set threading mode to enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6b50ecbae86642a66feed08ab563d2e79ecbac0"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:40:05.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50e1083f496bac168ebc6fb1e5e16d6d9573f8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/50e1083f496bac168ebc6fb1e5e16d6d9573f8db"}]},"branch":"refs/heads/master"},"d4d4f103d21804dec1c4789fa5354f95589dd61d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-24 21:39:37.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/3"}}},"commit":{"parents":[{"commit":"b6b50ecbae86642a66feed08ab563d2e79ecbac0","subject":"DNM: Set threading mode to enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6b50ecbae86642a66feed08ab563d2e79ecbac0"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 21:39:37.000000000","tz":0},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4d4f103d21804dec1c4789fa5354f95589dd61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d4d4f103d21804dec1c4789fa5354f95589dd61d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"341aa2650e222b9e2ec1be3dfbe7857af9779da9":{"kind":"REWORK","_number":4,"created":"2026-02-27 20:02:52.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/4"}}},"commit":{"parents":[{"commit":"fe918a221db6f87b73d79310c6ab33e6d06be8f2","subject":"DNM: Set threading mode to enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe918a221db6f87b73d79310c6ab33e6d06be8f2"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-27 20:02:27.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/341aa2650e222b9e2ec1be3dfbe7857af9779da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/341aa2650e222b9e2ec1be3dfbe7857af9779da9"}]},"branch":"refs/heads/master"},"841820d29948191fc77ce11a1e9406a4925b709c":{"kind":"REWORK","_number":5,"created":"2026-02-27 20:12:09.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/5"}}},"commit":{"parents":[{"commit":"fe918a221db6f87b73d79310c6ab33e6d06be8f2","subject":"DNM: Set threading mode to enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe918a221db6f87b73d79310c6ab33e6d06be8f2"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-27 20:11:40.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/841820d29948191fc77ce11a1e9406a4925b709c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/841820d29948191fc77ce11a1e9406a4925b709c"}]},"branch":"refs/heads/master"},"369252a1b3670694dd3e88e48fd96ae9af9222ad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-03-26 17:53:37.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/6"}}},"commit":{"parents":[{"commit":"0285079eddd62d2db9787ecc442922afe5e927ba","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0285079eddd62d2db9787ecc442922afe5e927ba"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-03-26 17:52:46.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/369252a1b3670694dd3e88e48fd96ae9af9222ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/369252a1b3670694dd3e88e48fd96ae9af9222ad"}]},"branch":"refs/heads/master"},"44726a05831e7476fc7e71f97c651bd38fa3dd2d":{"kind":"REWORK","_number":7,"created":"2026-03-26 18:46:11.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/7"}}},"commit":{"parents":[{"commit":"0285079eddd62d2db9787ecc442922afe5e927ba","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0285079eddd62d2db9787ecc442922afe5e927ba"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-03-26 18:37:53.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44726a05831e7476fc7e71f97c651bd38fa3dd2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44726a05831e7476fc7e71f97c651bd38fa3dd2d"}]},"branch":"refs/heads/master"},"28025464741a9df80dbde674cf3d02d22d94ecde":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-03-26 19:39:15.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/8"}}},"commit":{"parents":[{"commit":"4161843a5bc8790f289a30c910fa691bbb58d874","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4161843a5bc8790f289a30c910fa691bbb58d874"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-03-26 19:38:31.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28025464741a9df80dbde674cf3d02d22d94ecde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/28025464741a9df80dbde674cf3d02d22d94ecde"}]},"branch":"refs/heads/master"},"fe31e23c3bcdacb6eb7a4522462d7256140d50ff":{"kind":"REWORK","_number":9,"created":"2026-03-26 19:54:35.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/9"}}},"commit":{"parents":[{"commit":"4161843a5bc8790f289a30c910fa691bbb58d874","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4161843a5bc8790f289a30c910fa691bbb58d874"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-03-26 19:53:56.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe31e23c3bcdacb6eb7a4522462d7256140d50ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe31e23c3bcdacb6eb7a4522462d7256140d50ff"}]},"branch":"refs/heads/master"},"eec9131d47ffd3f8243cc7dcff3abf72d6b26c6b":{"kind":"REWORK","_number":10,"created":"2026-03-26 19:56:31.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/10"}}},"commit":{"parents":[{"commit":"4161843a5bc8790f289a30c910fa691bbb58d874","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4161843a5bc8790f289a30c910fa691bbb58d874"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-03-26 19:56:11.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eec9131d47ffd3f8243cc7dcff3abf72d6b26c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eec9131d47ffd3f8243cc7dcff3abf72d6b26c6b"}]},"branch":"refs/heads/master"},"362b984b3f5da1dc535da58d223522314a455982":{"kind":"REWORK","_number":11,"created":"2026-04-16 15:04:15.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/11"}}},"commit":{"parents":[{"commit":"1fcbe44bd21fdbff13d414771f7c05360b4c1358","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1fcbe44bd21fdbff13d414771f7c05360b4c1358"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-16 15:01:51.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/362b984b3f5da1dc535da58d223522314a455982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/362b984b3f5da1dc535da58d223522314a455982"}]},"branch":"refs/heads/master"},"6279ed48595736a7325ba28ad38432155a9b1381":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-04-16 18:28:13.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/12"}}},"commit":{"parents":[{"commit":"2a11366106c5dd6edc63e29aa4aebc2b21c1c435","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a11366106c5dd6edc63e29aa4aebc2b21c1c435"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-16 18:27:53.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6279ed48595736a7325ba28ad38432155a9b1381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6279ed48595736a7325ba28ad38432155a9b1381"}]},"branch":"refs/heads/master"},"2815fcfc79a02d220b4cd56cd8538e6567dca70a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-04-16 19:44:03.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/13"}}},"commit":{"parents":[{"commit":"44c1d20eef779111e64cc9c025f905ef67e08db5","subject":"Add zuul job for tox py312-threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44c1d20eef779111e64cc9c025f905ef67e08db5"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-16 19:43:40.000000000","tz":-180},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2815fcfc79a02d220b4cd56cd8538e6567dca70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2815fcfc79a02d220b4cd56cd8538e6567dca70a"}]},"branch":"refs/heads/master"},"05e9c5b5d70b00da14933a5770d072bcd9154372":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-06-01 18:16:28.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/14"}}},"commit":{"parents":[{"commit":"6c266262124fb16ed7c566b1e0674f8d38d3e706","subject":"Get unit tests to run (db fixes)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c266262124fb16ed7c566b1e0674f8d38d3e706"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-06-01 18:16:28.000000000","tz":0},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05e9c5b5d70b00da14933a5770d072bcd9154372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/05e9c5b5d70b00da14933a5770d072bcd9154372"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"2815fcfc79a02d220b4cd56cd8538e6567dca70a","theirs":"6c266262124fb16ed7c566b1e0674f8d38d3e706","contains_conflicts":false}},"e0a6f36f90ac70ae2bfb2dc20c5b950b11efdafd":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-06-10 12:50:16.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/15"}}},"commit":{"parents":[{"commit":"eb6f969c833e62a8ccfdad27ec6791ff4ebd302e","subject":"Get unit tests to run (db fixes)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb6f969c833e62a8ccfdad27ec6791ff4ebd302e"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-06-10 12:50:16.000000000","tz":0},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0a6f36f90ac70ae2bfb2dc20c5b950b11efdafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e0a6f36f90ac70ae2bfb2dc20c5b950b11efdafd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6c266262124fb16ed7c566b1e0674f8d38d3e706","ours":"05e9c5b5d70b00da14933a5770d072bcd9154372","theirs":"eb6f969c833e62a8ccfdad27ec6791ff4ebd302e","merge_strategy":"recursive","contains_conflicts":false}},"dbcaac281185b535cbb28ee7dad718d5baaec79f":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-06-10 15:30:58.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/16"}}},"commit":{"parents":[{"commit":"d078feefe59ede0148838f2d64eb21db878420fe","subject":"Get unit tests to run (db fixes)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d078feefe59ede0148838f2d64eb21db878420fe"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-06-10 15:30:58.000000000","tz":0},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbcaac281185b535cbb28ee7dad718d5baaec79f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbcaac281185b535cbb28ee7dad718d5baaec79f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"eb6f969c833e62a8ccfdad27ec6791ff4ebd302e","ours":"e0a6f36f90ac70ae2bfb2dc20c5b950b11efdafd","theirs":"d078feefe59ede0148838f2d64eb21db878420fe","merge_strategy":"recursive","contains_conflicts":false}},"f8702043344eb662c5b021f56ff9ec64b5936407":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-06-22 12:29:25.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/17"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-06-22 12:29:25.000000000","tz":0},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8702043344eb662c5b021f56ff9ec64b5936407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8702043344eb662c5b021f56ff9ec64b5936407"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d078feefe59ede0148838f2d64eb21db878420fe","ours":"dbcaac281185b535cbb28ee7dad718d5baaec79f","theirs":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","merge_strategy":"recursive","contains_conflicts":false}},"34c049bccdaf1aeed49d53207730b96e6ed5ff11":{"kind":"REWORK","_number":18,"created":"2026-06-23 12:38:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/18"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-06-23 12:38:14.000000000","tz":-240},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nThis patch and https://review.opendev.org/c/openstack/cinder/+/977883\nare mutually exclusive.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34c049bccdaf1aeed49d53207730b96e6ed5ff11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34c049bccdaf1aeed49d53207730b96e6ed5ff11"}]},"branch":"refs/heads/master"},"3ac169fcc66305414a0ccfe986346a6c3cdd999f":{"kind":"REWORK","_number":19,"created":"2026-07-09 14:59:37.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/19"}}},"commit":{"parents":[{"commit":"7db0225cc8e08477e56cb4e82fc2666033bbc2ae","subject":"Merge \"Add agentic coding guidance and docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7db0225cc8e08477e56cb4e82fc2666033bbc2ae"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-07-09 14:59:29.000000000","tz":-240},"subject":"WIP: Run Cinder\u0027s Backup service in native threads","message":"WIP: Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ac169fcc66305414a0ccfe986346a6c3cdd999f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ac169fcc66305414a0ccfe986346a6c3cdd999f"}]},"branch":"refs/heads/master"},"49489ad902c056c2916b0c2d4e478ea3b9566da0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":20,"created":"2026-07-28 19:27:30.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/51/977851/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/20"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-07-28 19:26:52.000000000","tz":-180},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49489ad902c056c2916b0c2d4e478ea3b9566da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49489ad902c056c2916b0c2d4e478ea3b9566da0"}]},"branch":"refs/heads/master"},"9fa5465226f8f9f9644fc7b90c5e6b8fe6a9a40c":{"kind":"REWORK","_number":21,"created":"2026-08-05 12:17:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/21"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-08-05 12:13:33.000000000","tz":-240},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fa5465226f8f9f9644fc7b90c5e6b8fe6a9a40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fa5465226f8f9f9644fc7b90c5e6b8fe6a9a40c"}]},"branch":"refs/heads/master"},"5018d7685933aa204b8c1c8e92b5f55a0d89ee6a":{"kind":"REWORK","_number":22,"created":"2026-08-05 12:22:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/977851/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/22"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-08-05 12:22:05.000000000","tz":-240},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5018d7685933aa204b8c1c8e92b5f55a0d89ee6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5018d7685933aa204b8c1c8e92b5f55a0d89ee6a"}]},"branch":"refs/heads/master"},"9ad99a1e66fbd3227f4b4c0673378758a8bd5896":{"kind":"REWORK","_number":23,"created":"2026-08-06 04:48:51.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/51/977851/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/23"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-08-05 22:33:00.000000000","tz":-240},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nco-authors: tests only\nCo-authored-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\nSigned-off-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ad99a1e66fbd3227f4b4c0673378758a8bd5896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ad99a1e66fbd3227f4b4c0673378758a8bd5896"}]},"branch":"refs/heads/master"},"75143be0b75a6df8ddcda2e0aa6bb5335ff0a96b":{"kind":"REWORK","_number":24,"created":"2026-08-06 16:33:11.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/51/977851/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/24"}}},"commit":{"parents":[{"commit":"c9291171cf1448ce5a887c513418772191f11123","subject":"Merge \"Storwize: stop tests spawning cinder-manage subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9291171cf1448ce5a887c513418772191f11123"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-08-06 16:32:31.000000000","tz":-240},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nco-authors: tests only\nCo-authored-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\nSigned-off-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75143be0b75a6df8ddcda2e0aa6bb5335ff0a96b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75143be0b75a6df8ddcda2e0aa6bb5335ff0a96b"}]},"branch":"refs/heads/master"},"c07c49c586d1667d3e11a7142689894fd9196c74":{"kind":"REWORK","_number":25,"created":"2026-08-10 20:49:40.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/51/977851/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/977851/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/977851/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/977851/25"}}},"commit":{"parents":[{"commit":"6fad01356c5950653b3734a762ed2386043579d2","subject":"Merge \"Fix: Unset default type for non-existent project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6fad01356c5950653b3734a762ed2386043579d2"}]}],"author":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-02-24 18:25:21.000000000","tz":-180},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-08-10 20:49:26.000000000","tz":-240},"subject":"Run Cinder\u0027s Backup service in native threads","message":"Run Cinder\u0027s Backup service in native threads\n\nHandles the backup service concurrency through Python threading\nprimitives if native threading mode is enabled, otherwise use\neventlet as usual.\n\nco-authors: tests only\nCo-authored-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I6320b495e7da077fb7204b50c8d06d736a36e946\nAssisted-by: claude-4.6-sonnet\nSigned-off-by: Fernando Ferraz \u003cfernandoperches@gmail.com\u003e\nSigned-off-by: Jon Bernard \u003cjobernar@redhat.com\u003e\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c07c49c586d1667d3e11a7142689894fd9196c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c07c49c586d1667d3e11a7142689894fd9196c74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
