)]}'
{"id":"openstack%2Fcinder~979100","triplet_id":"openstack%2Fcinder~master~Icbfbe6b88eb3003e4108c5d2482c665472363e7c","project":"openstack/cinder","branch":"master","attention_set":{"36737":{"account":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"last_update":"2026-06-19 19:55:54.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-28 00:41:16.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icbfbe6b88eb3003e4108c5d2482c665472363e7c","subject":"VolumeNumberWeigher: support cluster-aware volume counting","status":"NEW","created":"2026-03-06 04:38:56.000000000","updated":"2026-06-19 20:02:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"f42545a74ef88d105736a892f51ce8871bd38b07","_number":979100,"virtual_id_number":979100,"owner":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-06 09:13:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":-1,"date":"2026-05-06 12:25:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-06 04:42:26.000000000","updated_by":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"reviewer":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"state":"CC"},{"updated":"2026-03-06 04:44:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-06 04:48:28.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-06 04:49:27.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-06 04:51:43.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-06 04:54:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-06 04:57:05.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-06 05:13:43.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-06 05:18:10.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-03-06 07:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-06 10:29:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-06 21:55:52.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-09 01:20:01.000000000","updated_by":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"reviewer":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"state":"CC"},{"updated":"2026-04-23 06:27:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-28 00:48:20.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-04-28 02:02:29.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-04-28 03:01:52.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-04-28 04:51:59.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-06 12:25:41.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-06-19 19:55:54.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"6b2aeebed889b1ac4eef7a8032bddf45d996f26b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-03-06 04:38:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d25bbcd68d117a4ba1d01ab6988ca70ecbc3d827","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-06 04:42:26.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/979100 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"52bf466210d84fdd1d13892aa5083780d38dc8ab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-06 04:44:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2fecc9ae84d3499781dad6cc9d5f3c99 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a2b93f05a2104f6ebf51a0285aee5921 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":1},{"id":"d96f383a5f815aa8f9939da960c5ce58d64f0ae5","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-06 04:48:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/8bd0017d444647debe4c3a7ac3d62e15 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2720bfa301e229d652d1ef07f93f28c53419bde4","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-06 04:49:27.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/979100 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dce1070e878d42e477487833f827bb90aaa9b881","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-06 04:51:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/f49d53570e644d6faca93c214741bddb : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c1ac5ee1560c52c69bb11fbb82489613e50a0946","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 04:54:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe54713d4e741ba9ee0d42ddddcd409\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4633d389539c494083d612fa99ed6e9b : FAILURE in 6m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6fae377da73b4d418243243ba4b5f189 : RETRY_LIMIT in 3m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1e079d829db64ccd9273ce7603daf5e5 : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66a189b76e43ad48f1e429ecf84e9119da80c63a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-06 04:57:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b5f9d5bf517b4cfebd8df0868211bf2f : FAILURE in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6332648d372f54be20f1024438da52eefcf36f3b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-06 05:13:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":1},{"id":"5472783a6106de748d5fe4f7733a65c95c350499","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-06 05:18:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/979100 : FAILURE in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"712e5cffc2cc8f14ab48e576e709af989f3ac8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 07:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/728a0b47082c47c1a8c09913247f52d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d7cdcd55baf415d8bedb0500fbc2e98 : FAILURE in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e491cb386d8849f2a4796846440e4e44 : FAILURE in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93bc1f21487542ccbadde09225559fb5 : FAILURE in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a2b25c4ee0a4889b5b6ab3894a83e38 : FAILURE in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22541a76465414fa9c696d0b054517c : FAILURE in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3c805602c42347bfb92af5cf0504b1d0 : SUCCESS in 30m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1eb6c6b66ceb4fbd9f4035d1261e2aa0 : SUCCESS in 37m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa263d03fb1a4e5d84dfc6ff87553656 : SUCCESS in 1h 33m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78bb5f6fccdd4c79b2936b923674957d : SUCCESS in 36m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/667936677b2d42fd8e3e0340a627b473 : FAILURE in 5m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9d6cf61a29134cc1adf379deb4edc0b8 : FAILURE in 2m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/56377a2f8dcd4fc381241612197cdd8e : FAILURE in 4m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7e4f63c044e44959390d1fb1aefecc8 : FAILURE in 4m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/21270b3c98be4bba9895dde9a1d02532 : FAILURE in 3m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a59f965ec83c4303bd83ea4b25841ee0 : FAILURE in 40m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08f57871cca04efe9a8514b38cca0e33 : FAILURE in 4m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab1f3e9ade1c42fa97d6ecc4b6f623be : SUCCESS in 1h 48m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8df482a6052c4462ae608f20427a7b05 : FAILURE in 1h 43m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0eed6b2feabd416b8314649c0fd6437a : SUCCESS in 2h 12m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e9ac254166d54b08adfc18decbc18700 : SUCCESS in 1h 24m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/33420efe1b8c4464b9f4f807720e5cbc : SUCCESS in 40m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4a2e638530f4d1e91a635806ca8a34c : SUCCESS in 50m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3946ac197cb44042ba5a3d4c2c95b456 : SUCCESS in 1h 18m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ee2eb47c70ff4e62a066860bdb3605a6 : SUCCESS in 45m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79b5f3898ccf46a2a0c9fe322b901b54 : SUCCESS in 59m 10s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/9b501caafc974d349174806db6732b40 : FAILURE in 2h 41m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dd2afeba761a42d5aa92143c0d21109d : SUCCESS in 1h 40m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455078caa95442069cfb391b2d09bcc7 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a7f793a9147fc8588cf6b38389abd3ee1b8d61d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-03-06 08:01:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"66dcb941b97a75ac73f87fefd183b14a71d613c2","author":{"_account_id":38833,"name":"maruthi","display_name":"HPE 3PAR Storage iSCSI CI","email":"maruthi-kumar.pabbisetty@hpe.com","username":"mapabbis"},"date":"2026-03-06 08:04:57.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/979100 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d72c62cf6cbc32775826884195330f23d36cc967","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-06 08:07:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/87ab6a68dfe14600a550ca2374cf73fb : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/0c5cbfa4ee0c4d07bb61bfebd138f3a9 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":2},{"id":"c2967db010939e55962ae484453d8147e7c4663e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-06 08:11:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/75860d52a14c495cae73d6e769924b45 : FAILURE in 9m 30s","accounts_in_message":[],"_revision_number":2},{"id":"52457e08c681526ef41dde872961f5506b3380f5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-06 08:12:38.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/979100 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ca716506520f7162b9d26bceb2759635202c8e0b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-06 08:13:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/dfafac18235b4cd8a9a2cb46d569269d : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b435f7281f767df0de0e6e58116bed6d0ee2b477","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 08:18:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9fb802e5e2d4ad79dd4c7855d570cf5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d055fd1c857841b9bdd6fae33e7f2a48 : FAILURE in 7m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/77ae30f1256e4349a55ec0f5662f4596 : RETRY_LIMIT in 3m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ed468b0df67b4dfcb28a5b18b2f508f4 : FAILURE in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0063e50b1cda02259745fd503ab356d24406348e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-06 08:19:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/59750f6159c446ea95b35ca0ae9d8fee : FAILURE in 17m 29s","accounts_in_message":[],"_revision_number":2},{"id":"20e09254718553cb9794f27ec5bdbd487a41428f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-06 08:38:30.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":2},{"id":"dfb015120056528acbd60256a4878a193ee37135","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-06 09:06:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/979100 : FAILURE in 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"283a5ed88782229f0d40542e6c1cde08fbf13331","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 10:29:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3184/.","accounts_in_message":[],"_revision_number":1},{"id":"861e2e78389a3764efaf5ef2be16459c34878eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 10:32:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86d95bb525a14a16954f89593e3191a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6e168819f14ba49465bdee82532f06 : FAILURE in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0a94e14d6384900961f99739c0f5629 : FAILURE in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fece36f5c02e4b4dbe990ab96b17acb1 : FAILURE in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ce859f6f09b4744b2f7c5680e07b1f1 : FAILURE in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1770f7b3c644f9682e78acce70cf99a : FAILURE in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/706ab4803db6413ab7e8852cbca71e8f : SUCCESS in 30m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78a1c30a2e1e4686b80323ca332f4c8e : SUCCESS in 30m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36f4ad654fab487086e3d2de3c8c8869 : SUCCESS in 1h 42m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d19b5542f8241129382d4f43bf44fd1 : SUCCESS in 1h 04m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c47ef83f4a44feb92f80cce4a3092fa : FAILURE in 4m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9b6d4ecbb30d472c9d85b1fea682ca5f : FAILURE in 2m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e97d69cdc08e4070a4419c9bbec6e57c : FAILURE in 4m 08s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b51814533389467f837c5106906c6211 : FAILURE in 4m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/397cbe03edae4ee58b8383619a55f2c0 : FAILURE in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/92b99e46e4ac41489a6ad45c69af580f : FAILURE in 40m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c68ce685702400b87bd065c3fc75354 : FAILURE in 4m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e45936ce508f4caf918d9af3405ff55e : SUCCESS in 1h 14m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e0bc41a76c95433a8a64f82fcb11a3ba : FAILURE in 2h 29m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8dd9c94405c345d6aac840837ac58ace : SUCCESS in 1h 43m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d7c58d4a57ad49d8804a5cdcc70bd0b0 : SUCCESS in 1h 50m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/912259163e474ac2b6fa238a93e9c6de : SUCCESS in 25m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60915ee1519b485ba7910880e13c411b : SUCCESS in 40m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a155ee549322477da96c2f69f75ae328 : SUCCESS in 1h 37m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f5bcfe8dbae847928279d6a1951cacce : SUCCESS in 41m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9ee4df3ae4f4968b31c034a229d1046 : SUCCESS in 2h 17m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/419c5c5abf7e43879eec6c6afce83ce6 : SUCCESS in 2h 04m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d98fec563ea34e488f22938418e49b43 : SUCCESS in 1h 58m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e615fab1f42944599fc0951be1f9ea8e : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"520934fcfcee0c79573d9e19fbe9664706dca741","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 12:23:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3357/.","accounts_in_message":[],"_revision_number":1},{"id":"54b727cf75fd0c9cbf3eb03681b4087090d937e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 17:48:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3190/.","accounts_in_message":[],"_revision_number":2},{"id":"00166622d7a2e9235f107864fc540b07737618db","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 19:16:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3363/.","accounts_in_message":[],"_revision_number":2},{"id":"218461d8617c36e9eaff8d6de93688fcd7966215","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-06 21:55:52.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_983/index.html : SUCCESS in 97m","accounts_in_message":[],"_revision_number":1},{"id":"2afb692fa68648c63922fabe4388e0824fb33d02","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-07 02:49:31.000000000","message":"Patch Set 2:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_987/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":2},{"id":"dd16367ff0bef0d9bf22e5f8c2a9930e477f6b21","author":{"_account_id":36686,"name":"Inyong Hong","display_name":"hongp","email":"inyong.hong@samsung.com","username":"hong-p"},"date":"2026-03-09 01:20:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f88e1d55b2c775e62ffc1bf030d6459b7b5f9f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-15 05:10:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b3cf15e01e10f82b76bfe7113f4ef17a390e7431","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:11:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/566d64409df64c13bf5e998a10fef538 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/00763efecd594cdfa2d70cff3f5d10ed : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"2854ae47e7a206460b41d779eb3e1d194d9c588b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-15 05:14:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1765349fcef05b283e298265998a43ee801a050","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:15:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e6ca12f129ac4185a826ad86b720fa33 : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/6860cf2f22cf492586b83f5ba53e0afc : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"bc766a4c3b3b1f2250d4a7e8beb06caec7e6f72c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-15 05:17:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e80b538c3d662acfb5e080fd2070d80e83441603","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:19:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/088eea36f91149a29f937f0007810cac : NODE_FAILURE in 0s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9012d8c7c1664c2cbca4e7b9f70744d8 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a4441c74425669224e117b038ce1ee7433509ca1","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-15 05:44:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e32663cdb0b7463f9934f16b7d62469f : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":5},{"id":"f7409fac807d5273d0b0398fb4aa8a1d1917b3d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 05:46:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e7a42acaec4e6bb29b96bd950651c8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/882c0ee9ce4e4e69a8b8e725927dd91b : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6df54816c77c4ddd8c91e1b63b448d1a : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bca8c1381b854ceca4b1c5e83042db14 : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4cc857c191c77f82307478f767fc1dc04e8d2d1e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:48:54.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":3},{"id":"17af9e6521e1f03748059fa7fa2f8750828103d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:53:25.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":4},{"id":"5dd263aa3b64a581ef3c1b1ca0b45aa6039a4c9f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-15 05:55:41.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":5},{"id":"71114c6b0c2a83ee6038866f2e38ccc55d8679bb","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-15 05:57:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/cdb72ca720a54f608f1fe57f0db9d38f : SUCCESS in 37m 56s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"850de961f70a153bcedc5590fd92f099dd8dfec1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-15 06:02:11.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/979100 : FAILURE in 39m 37s","accounts_in_message":[],"_revision_number":5},{"id":"736d45f7271fca58b357046662ef1fc1f575af22","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-15 06:05:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a4c8601998b64c5791fe1462944602cd : SUCCESS in 45m 19s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"1fbf6b83c3e3747190827af3ba0eeef8a9ae7a41","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 06:42:33.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3636/.","accounts_in_message":[],"_revision_number":3},{"id":"3e4c3421c9a68e71709c52f0865da09715b57993","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 07:47:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3637/.","accounts_in_message":[],"_revision_number":4},{"id":"94fedca3e19ca54c6b02638eb61305cc5951eb36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 08:33:06.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2f90a044df1482c9251cec76015484f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973b6a98503a40089d0cea6e6de958de : FAILURE in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0696102729b45bdb94a0702b2d17941 : SUCCESS in 7m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9de64931f9d455a96727c5b240724c6 : SUCCESS in 10m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eec7b4d9628f49a0820523aa14df18e5 : SUCCESS in 14m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a930c58649e48e99df6f74a9d83fb28 : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/64f9b598fb084eb498b0f8b31ec76308 : SUCCESS in 1h 04m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5767c4ba49bc48e9a169814d5856a606 : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76cff2d16b1c4c049f732d903b17dc7c : SUCCESS in 1h 44m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5293d99d4f0646348c7171fa5cdce84f : SUCCESS in 33m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fbc39a0f1169403fa983f697fef746a4 : SUCCESS in 21m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/28132d215690458a82a93c8759c2fa82 : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b18ebc41c23d43e487b6ed30058752d7 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d95c0f31226499cab33e44ba66ba55b : SUCCESS in 5m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f3bbc64bba494487bb11baa578f3cd2e : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7dd7cd521db412eb69f625df3a5ae7f : FAILURE in 43m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b03a19a94d094829b7297c378f93cc66 : SUCCESS in 4m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1dc888c7b0874b259621aa09d775d789 : SUCCESS in 1h 50m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c2ab07594e24dcf96b06cf174de29ab : FAILURE in 1h 24m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8c62f58f24ee42b0b2885f1172c19ba1 : SUCCESS in 1h 52m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6342ef82473d4b0aa4c0b216c62cf227 : SUCCESS in 1h 46m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b9735692ec14463c91ffaf126925278c : SUCCESS in 17m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0805e1773a64676b24a1cdbdf1a4a85 : SUCCESS in 1h 19m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/76c927093386474796d279fc552d877a : SUCCESS in 1h 26m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c52172147a1e4777b49506eb72535cf4 : SUCCESS in 43m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99256066c3ca4bca91a5fff9956f021d : TIMED_OUT in 3h 10m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/312e67cda0474453879e7a38e810dda6 : SUCCESS in 51m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ceef3c67747459d970bf033b874a9c2 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":5},{"id":"e7be73499313feb8e407223162b9613aa7650e44","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 09:01:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3638/.","accounts_in_message":[],"_revision_number":5},{"id":"9b95c75547c73c505a26e3cc74ef0e4e16691bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-16 00:32:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"143983e065972395aa17663c13bfda5515302313","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 00:38:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b3468e46d77943578d887ff4e69c3510 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1cf984c68d9f4b25a8ea6dcd4fa79405 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":6},{"id":"383d4149fff742aab39df18839390ed6e89aa82f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 01:01:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2e2a4578ac482db73cf8f71af3d003\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9be78711fa1048669119437f27301893 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e38cffbfae87445e8e7bcec753276989 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/52af1d6b18e644419b7f425f9b03d798 : SUCCESS in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6db501809e4fcb3c84e50edbe7ef61ebb77fc2b8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 01:10:27.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":6},{"id":"81c4256d59f3223328d35c6399356b5f16532f3e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-16 01:12:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver build/776b4bb018af47a79d3c7c9200e1fc9c : SUCCESS in 38m 57s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"cf5a16f7dc077284407df3345faa24d8727945d2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-16 01:14:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/979100 : FAILURE in 39m 53s","accounts_in_message":[],"_revision_number":6},{"id":"da44c609ea6819c8bd1395bfbec43b4944c78479","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-16 01:15:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/11b6d455349840fba5d244a069cf7fa7 : FAILURE in 41m 44s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"79334f76cfd8ca8eea3e16f0e44f89ef6745aec4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-16 01:19:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/1e49a8469a7443f48e03d1d15902aab8 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":6},{"id":"aa6aa6d5d353b78560df8407b407360441107091","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 02:12:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3812/.","accounts_in_message":[],"_revision_number":3},{"id":"2d1f04f2df408d855f23ea161bcd6efbd9512408","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 03:13:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3813/.","accounts_in_message":[],"_revision_number":4},{"id":"0b51247606b0115fa18b25bd7dd5e9079c9fd3d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 03:40:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3654/.","accounts_in_message":[],"_revision_number":6},{"id":"a1b405c27f39b2ed8cc584a8fcf081c202708d22","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 04:32:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3814/.","accounts_in_message":[],"_revision_number":5},{"id":"02e4fe7b7f8289ea6e8a76e6edde6ca29ed37212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 06:25:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f7f3a8aec94bf2b3ce2c7a5558ebc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de966b489434aad8149480a41b8e35a : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b44b475a84844a6c861f2689cb72234e : SUCCESS in 10m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e35d56a1ed948b991d17fab99b7a8cb : SUCCESS in 10m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0da3415893e1481ebc6850ee053da84b : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b463addae9a541e68c7839f9de718ce1 : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/4b363ffb884245a79122953718819056 : SUCCESS in 33m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ef0a3ace3a34e79b041b523cf30052e : SUCCESS in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3c46fe9aa6234ae2a3ee20696e1560df : SUCCESS in 1h 20m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60d9080363e646e79a63c49c051fd763 : TIMED_OUT in 2h 33m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ba21bd6ea0e47708509f7907100eb1e : FAILURE in 25m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c4a09d2653724e509a36d88ed48e948d : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1fa01149b32444e399384be225b5e239 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/442a4665686741489bcbe271ac5f3713 : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6b9a893f2f5848ba8ee114dec8558a0e : SUCCESS in 6m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cbe41f87c2404746b162e07ab26a70a6 : SUCCESS in 39m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/69a8e4f37ba74240bb1519cee83281cf : SUCCESS in 4m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b26e0f702644415aa31993b8af4c51ec : SUCCESS in 1h 59m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4657cf340a1e40bc9673261604e798ff : FAILURE in 2h 01m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/747271eb27574e28867b037188579208 : SUCCESS in 1h 27m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e9cbfbfd982547c1a55ecf33c6c1f37f : SUCCESS in 2h 04m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fb134c433e6343f0acec57675a7978c2 : SUCCESS in 26m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cf5b0ccdbf3b4bc4b3b624a60a9ddd82 : SUCCESS in 40m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2d0921de53a643938fdaa41d2122baab : SUCCESS in 1h 32m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51213c811bb447f38e6aa463d1d6a94b : SUCCESS in 22m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dffc7142da6b417680346ae8b7e6169b : SUCCESS in 2h 17m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebe9e2c721634f31ada98ee73a380fa6 : SUCCESS in 1h 19m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e265cf644f964dc8ae57c3c3f6659ddf : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":6},{"id":"4b82494e2250184d5fa028b5847cc313455097b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 23:58:39.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3830/.","accounts_in_message":[],"_revision_number":6},{"id":"5e76176c700e61056258c32710b1ddc4de7a103a","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-17 02:14:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6eedcc9193aab41219404a019282661a5f88583d","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-20 05:34:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c7923d2ed74e397716d8aa5be91938f6438554ae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 05:40:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d5516cd13caf49b3bced3222e8c652f1 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b3451f6ecdc1411f9698f8f6e35b0a81 : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":6},{"id":"6259e063aa974e648582b9b2d9dd728c43c24b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 06:17:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/990a3ec9c4684f3ab7988df01082db2a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd7b8f1aea024439ada5559b35fd6c16 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd34bf685f2c470d9b40ef2a27910500 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c56f66a693594a588b5c67bc7c44357d : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"587f8fc0fcdecb5b80df0195c53fc71c3a2e9607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 09:21:07.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53ea9ba551744b34ad17c4a3c2343ef6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8d0d3cf00448fe9acef073e5feedd5 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4597dec2d5ea4bed88a9cd592cf81c64 : SUCCESS in 8m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9d06d5829264b45968c4102c75d63d1 : SUCCESS in 10m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3273aacafff4203930ca54a776784c1 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800d5aa2c737477d9da5cd7abb83753f : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/cd814e22b6724a64b347d918d0dfdd3a : SUCCESS in 1h 00m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ff22d79deed422ca8e69d47c7260a70 : SUCCESS in 1h 04m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1888d8232ea471caf8cfc511320e5b1 : SUCCESS in 1h 08m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c751d01848414b55977cec1c59c85c6d : SUCCESS in 1h 05m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f0c2e98c027e4ca0a03cd5e88c4470ce : SUCCESS in 23m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fb01150d23b45db9af3e8f440fbccb9 : SUCCESS in 3m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/415cd9e5b47d4ce7a82d579c35558aef : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53407d86a40d43d387c5235d1e32f78f : SUCCESS in 5m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67f7c9b9ad4b4ccba40a5bbf09dbf70b : SUCCESS in 6m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bdaf2da1b11741abb8cce695593e1957 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60b0e2d461cc416d92ce6bdaa5d13690 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59d144ff00e44e59a4a287a2d7a612f6 : FAILURE in 1h 23m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2b88338e1f0449c4a5caf7c042687c66 : TIMED_OUT in 3h 38m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/04ef9bc4d83948a39a75e9c6d9c614c8 : SUCCESS in 1h 48m 43s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f186be2cbb4c4530939b2a6636ea3db2 : SUCCESS in 2h 00m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0d08fd4a2acf4f7c82b8c1046c2ddb4b : SUCCESS in 40m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f881637f732b41a482c74a8c1d9edabd : SUCCESS in 1h 03m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/05b1480e0dc44c43bd5cde7016f8ca80 : SUCCESS in 1h 44m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/efc7c9ac37b045d39e4dcf97811c69c9 : SUCCESS in 48m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc53063382334d6c886b3b7de1b400c6 : SUCCESS in 1h 19m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8f8ad4b50544b8882d0e64844f4930d : SUCCESS in 1h 48m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e37efedcd9a47a5a78006473ef91606 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":6},{"id":"416484a8cafbe39de21e16de1d2d2bc92c0e8436","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-23 06:27:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"41c842fefc7f4cc1f242ea979a0cca3ea62f4508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-04-28 00:41:16.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a7fa1b20ec8d12aae056891a310abc07740bbec2","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-28 00:48:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/76c0f38bdeac49b6b8715a09865a5f94 : RETRY_LIMIT in 12s","accounts_in_message":[],"_revision_number":7},{"id":"f3bc89c2b177fbc47f709388e2d1bc5131efe79c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 01:01:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53881993439a4611a0e266101691ada8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c509c6c235549768b5737538300528c : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cf9268a8ddb4115b0c59aca3a8d236b : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/463b7f3505714ebfb194927be598faac : FAILURE in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"41ca987c1920920711dd7de282c4df2de192c626","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-28 01:22:54.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":7},{"id":"25d7a4874829a6c2ff13ba396f5182c897bf4bf7","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-28 01:47:25.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4cffd08db7f04806a18b9398f7151599 : FAILURE in 38m 49s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"80b80ac651e099f999ef8c4de0ab53122834397a","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-28 02:02:29.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/979100 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"753a75a8b52f524b09aa9eff1da92efae7ca0afd","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-28 02:10:46.000000000","message":"Patch Set 7:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/979100 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"c1a829590e6f580450dacbf51766bce85ec4e814","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-28 02:11:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3e5310da0f1749b69f153165ef7c87e2 : FAILURE in 44m 31s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"4d52bc1bda3e3cb460cb394acdcaea2fff45b606","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-28 02:58:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3805/.","accounts_in_message":[],"_revision_number":7},{"id":"8acab3e6ad01dd88bcebcb0583c9b64b9ec3cd0e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-28 02:58:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3981/.","accounts_in_message":[],"_revision_number":7},{"id":"a6efe642d3a4d50999c57678b53b8f6418c78692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 03:01:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fd56f3ad5c4ba8b2bbf09a94a38b43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ede201b95be4e3aae8b543a97adef77 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08dbdd3f6f1343189d2b2ccc0729a1eb : SUCCESS in 11m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c358ee1cfa84fdc851a12b85571fa16 : SUCCESS in 10m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4dcccadd4ab4b85bcb5c901160a4627 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c5f4277945450ea50bcdc3956a68bc : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a6176a99abdc47659f86e9ccacc1ca76 : SUCCESS in 33m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4542152d30f4d19abaaa61d1cc42240 : SUCCESS in 1h 10m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f241c056c3834af886fc80189a91a9dd : SUCCESS in 1h 40m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c991841ff74f400cbf31235413e5862b : SUCCESS in 1h 01m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f2ba5a672a0f47438c0d3a4976eb0455 : FAILURE in 26m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/15b292f1d2d74cf0b296829e89b35364 : SUCCESS in 4m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/767a9b373b1544799f79b7e645d45321 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01dd567f05284495b965240af8981f19 : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e5dddc4661240a282b46b9b3827da33 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3661b1747c3c4c2fa8232c9d5ccb17d5 : SUCCESS in 37m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4658eaab3a94415ea3157ac0bfe6cde1 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5874a37418164132bc3520b4a9fd43bf : SUCCESS in 1h 44m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14e6df9f18104324bb9537201f0500b6 : FAILURE in 1h 24m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a53a91e9a8d4505af69bf771b6e7a89 : SUCCESS in 1h 56m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c6ae682258fb4af0a7f5e3847dca86a2 : FAILURE in 1h 29m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/48ad1f0b24dd42dcbd167c3581c0e779 : SUCCESS in 18m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a5295c5bcfe545d4bdc98ca252778f33 : SUCCESS in 1h 12m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c41fabfa549242d78af499d494fd1bf6 : SUCCESS in 1h 49m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d0287727be4f4dfe91d6ef680693c82b : SUCCESS in 24m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/450b4032ea9b4fda8e472388788dab96 : SUCCESS in 2h 14m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d07f4d3c67e4b1d832159852fe9e711 : SUCCESS in 52m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e36c6e8d1d0420d8b85845ab3fc95f2 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":7},{"id":"8ab0bb435369767c6d4a6507002a4f5eb556dee2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-28 03:08:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/979100 : FAILURE in 1h 58m 24s","accounts_in_message":[],"_revision_number":7},{"id":"525a21111ddf520e5d45d2e27af328123e01f634","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-28 03:11:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_04_2026_03_04_51\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"1a6e9d28fe5cc87b22be7269b929b6ec35a04d68","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-04-28 04:51:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-00-979100-7-fc : SUCCESS in 1h 4m 16s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-00-979100-7-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"13f8b5007aa00ed9c5a49c37f04d684c04b6e4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-05-06 06:35:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"8b0381ad6f2d035a29e081095e46ce3990c8336e","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-06 06:39:09.000000000","message":"Patch Set 8:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/979100 : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"57aa2a230260ed90ea89bd2230bf3431cbbeceeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 06:51:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c49a80a9924588bf24dc4a6341b8fc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da302b6566bb4636b59d3c635dd14adc : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/278de02a2de048eeaf8d03d6f6467654 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e37464f4c856412c8f7eceff7654ba00 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f738ac2224ea3a81b965b9c8b9bd1f699d55af46","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-06 07:10:57.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/d042041203a74ca5b0a2bbd9150a168b : FAILURE in 35m 11s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"7b951f09916fb0b6a82aee6b75a2c0fc7ccc27cd","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-06 07:20:52.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/979100 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"bbb35ea702d9b75f52f021ec97f46b65d414b97d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-06 07:42:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4080/.","accounts_in_message":[],"_revision_number":8},{"id":"7d9b7bb8958faa71d050a9b5b69f165b1fc7cefb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-06 07:42:33.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Icbfbe6b88eb3003e4108c5d2482c665472363e7c .","accounts_in_message":[],"_revision_number":8},{"id":"c7a318a04f2fa2c208f66e4f8f4ec7c5cafa1a81","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-06 07:57:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3904/.","accounts_in_message":[],"_revision_number":8},{"id":"ab0911e6e5b37d655fd627d470dffab683054159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 09:13:38.000000000","message":"Patch Set 8: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/254d04d4d2134271a1296014ca43fbc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2b48499b6d4f45856f3770348b4637 : FAILURE in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d229c0a3fcb44fbd9156c562cd576478 : SUCCESS in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ed26d85ef83495f8d939c78af8bdcd6 : SUCCESS in 8m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2183339357a4ce29ece4d127f91fcad : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2817a7ea1741049ddbff7db6c7fd3c : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d1f0018e7e1a4acf9c2ab940580a4758 : SUCCESS in 1h 09m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70399ed9ce824081a1122f5a272a94b7 : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a73c1e11dcba49dd85064899d9016d89 : SUCCESS in 1h 31m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc70327014544bbb882c10ced8b0f634 : SUCCESS in 42m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/affb9d5d47cf4387aada37dda75ffb01 : SUCCESS in 20m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9bb587725b23460292fd972cfaef038f : SUCCESS in 19m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a60e53efe46e4b4fb16427cf874d8e6a : SUCCESS in 4m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9798817138a749e2b25c61c1f50df325 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2de450a3428044fb953ebc63486e91dc : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/451f1905d7ff4dfc95cf2f8088e4e37c : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b29ad29f7a374a14999d32aa49d92174 : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/311a51e464334a34bd13c1c4202050fe : FAILURE in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcfbddc604624f4c8fc2ea0a076c92fe : SUCCESS in 2h 04m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ffc9a0e64551462583714c9ad4a3aeb0 : FAILURE in 1h 50m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0e1d4431f98740a2b64e15c0a7b6afdd : SUCCESS in 2h 31m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2b630b0a945d42378f6ea6e7c2d65741 : SUCCESS in 1h 34m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6cd703be0d7545d2826552013cc705f1 : SUCCESS in 36m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3456c1b4bffe4d24b39273f3a03475e2 : SUCCESS in 1h 13m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b05301d56cca4cfcb0359c95b0e904be : SUCCESS in 1h 51m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5873e39577d94ef39463e37aeae5d5b8 : SUCCESS in 31m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32a421c1a25742de9935cc9e7d057789 : SUCCESS in 1h 56m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5434b987a05d4b67b8aee287dc1a67a3 : SUCCESS in 1h 20m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a46e296d3564fb69b00f84cd4f0a347 : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":8},{"id":"001d8740c70435a873950bed89d8521c6c77a343","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-06 12:25:41.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26f3943dd7f164d97a971f1a5a0075009b0b8932","author":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"date":"2026-05-07 02:05:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9ec08936b8e428320c0d6d297bfda51719d10164","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 01:14:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2b6e1455252146d88726fe85e93e7d2b : FAILURE in 28m 56s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/979100/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"4bc59245cf65dc9a506d09b094b9a2e34c485327","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:54.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5},{"id":"356273d51905208aae10692df70fce386f01310b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:13.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6},{"id":"d196b91f69b2e6817e9d98603dcca1e0a180fcae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:32.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"3679bf8150f6b5f5cf8173dab6b549020eaccd33","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:20.000000000","message":"Patch Set 8:\n\nBuild Failed","accounts_in_message":[],"_revision_number":8},{"id":"b0f0d9e40697ab025d8a3fe8b3e57ecc11c93344","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:54.000000000","message":"Patch Set 7:\n\nBuild Failed","accounts_in_message":[],"_revision_number":7},{"id":"f42545a74ef88d105736a892f51ce8871bd38b07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:17.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":8,"current_revision":"5b7968f929104bc9342790e5d7306cfae1120210","revisions":{"2c9ddbbb03be837e402a3b84ef999a30602262e0":{"kind":"REWORK","_number":1,"created":"2026-03-06 04:38:56.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/1"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:38:04.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c9ddbbb03be837e402a3b84ef999a30602262e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c9ddbbb03be837e402a3b84ef999a30602262e0"}]},"branch":"refs/heads/master"},"d2e0aa4baf56935882a44fb7cdf1e8916de8ae27":{"kind":"REWORK","_number":2,"created":"2026-03-06 08:01:28.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/2"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 08:00:47.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2e0aa4baf56935882a44fb7cdf1e8916de8ae27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2e0aa4baf56935882a44fb7cdf1e8916de8ae27"}]},"branch":"refs/heads/master"},"de615d8b03360f3103c25caa27e07f782f1554a5":{"kind":"REWORK","_number":3,"created":"2026-04-15 05:10:04.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/3"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-04-15 05:08:54.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de615d8b03360f3103c25caa27e07f782f1554a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de615d8b03360f3103c25caa27e07f782f1554a5"}]},"branch":"refs/heads/master"},"ac432efc3097e59fa055f1b6452d89ae2f799720":{"kind":"REWORK","_number":4,"created":"2026-04-15 05:14:19.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/4"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-04-15 05:13:58.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac432efc3097e59fa055f1b6452d89ae2f799720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac432efc3097e59fa055f1b6452d89ae2f799720"}]},"branch":"refs/heads/master"},"9ead820ab677fcd10cc350ab5d0cceeecc85d595":{"kind":"REWORK","_number":5,"created":"2026-04-15 05:17:47.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/5"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-04-15 05:17:23.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ead820ab677fcd10cc350ab5d0cceeecc85d595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ead820ab677fcd10cc350ab5d0cceeecc85d595"}]},"branch":"refs/heads/master"},"47e6172ffc82d501f1dc9df6c9b1578f0cee47c9":{"kind":"REWORK","_number":6,"created":"2026-04-16 00:32:52.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/6"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-04-16 00:32:26.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47e6172ffc82d501f1dc9df6c9b1578f0cee47c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47e6172ffc82d501f1dc9df6c9b1578f0cee47c9"}]},"branch":"refs/heads/master"},"d3e5262227d1fad3d81c9b5eaa78618026a1c330":{"kind":"REWORK","_number":7,"created":"2026-04-28 00:41:16.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/7"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-04-28 00:40:38.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3e5262227d1fad3d81c9b5eaa78618026a1c330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3e5262227d1fad3d81c9b5eaa78618026a1c330"}]},"branch":"refs/heads/master"},"5b7968f929104bc9342790e5d7306cfae1120210":{"kind":"REWORK","_number":8,"created":"2026-05-06 06:35:14.000000000","uploader":{"_account_id":36737,"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","username":"ohjiwooo"},"ref":"refs/changes/00/979100/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/979100/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/979100/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/979100/8"}}},"commit":{"parents":[{"commit":"1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1","subject":"Doc: Documentation updates for ibm-svf-volumegroup feature.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1aabcf92dbb8b56e05bf44f52cf8baac54e1dfd1"}]}],"author":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-03-06 04:34:01.000000000","tz":540},"committer":{"name":"ohjiwoo","email":"jiwooo.oh@samsung.com","date":"2026-05-06 06:34:55.000000000","tz":540},"subject":"VolumeNumberWeigher: support cluster-aware volume counting","message":"VolumeNumberWeigher: support cluster-aware volume counting\n\nUse cluster_name instead of host when counting volumes in\ncluster environments. Fall back to host in non-cluster setups.\n\nCloses-Bug: #2143374\nChange-Id: Icbfbe6b88eb3003e4108c5d2482c665472363e7c\nSigned-off-by: ohjiwoo \u003cjiwooo.oh@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b7968f929104bc9342790e5d7306cfae1120210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b7968f929104bc9342790e5d7306cfae1120210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
