)]}'
{"id":"openstack%2Fcinder~980848","triplet_id":"openstack%2Fcinder~master~I9a39bda816e64e43741bca75be723f4f505d0ea9","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-03-18 02:20:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a39bda816e64e43741bca75be723f4f505d0ea9","subject":"[Hitachi] Update docs for VSP One Block High End.","status":"MERGED","created":"2026-03-17 06:02:09.000000000","updated":"2026-03-18 04:08:52.000000000","submitted":"2026-03-18 02:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980848","meta_rev_id":"9110529574800cf272fed2a1eaf9985ab888bb29","_number":980848,"virtual_id_number":980848,"owner":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 02:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-03-18 00:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-03-18 00:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 06:08:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-17 06:40:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-17 06:41:08.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-17 06:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-17 08:08:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-17 08:20:52.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-17 08:37:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-03-17 08:42:40.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-17 08:46:30.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-03-17 09:23:52.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-17 10:27:44.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-17 10:29:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-18 00:24:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-03-18 02:58:42.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-03-18 04:08:52.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"148957f19cc16ec7b0afdbd0e57c00e3a52facae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-17 06:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc32eca1b7caef8ed3fa2da37fb5fa8b967877d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 06:08:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/67a44036856a460e82a61955267c29f9 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ad8c5f8844c4464b86a33eb308f22021 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":1},{"id":"4d7b3012da6eb142f343682ce49acc5d1d6e38f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-17 06:22:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b371699c012610e3cf4e0b3e4a03e187ded0eca8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-17 06:22:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"905dd26ee63089c2f852c3cea608ee2ae9b99ac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-17 06:23:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4135689ab65a40c9f2d4c3c35c6b98823bf28c6e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 06:29:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/31179dcdc8e248a3813c69f452a8c82f : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/5722b8e9aab0417182bf8303333313da : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":3},{"id":"90ac4fae61ba41e42a1c8cbfce3b3b547537cbd8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 06:40:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9a39bda816e64e43741bca75be723f4f505d0ea9 .","accounts_in_message":[],"_revision_number":1},{"id":"864d3064b6bd26ea6d3a1e21299682fdc1d5ffea","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-17 06:41:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/32b5969ad7bd448bb3040859fe27d84a : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ab4e5bcaedd9da54e7d3c3f1d174f6da5396a9c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 06:47:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5a4bcacf4c4e2586e77089b515fc39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83a84909ed384585a3c564ddc6670bba : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e8dfff712914ee2b57bbe67e09633c5 : SUCCESS in 15m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4bddb49480be4442932f02fe08ed5076 : SUCCESS in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ad1e9856e063c1badd2bdd2804839be581ba83c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 07:01:32.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9a39bda816e64e43741bca75be723f4f505d0ea9 .","accounts_in_message":[],"_revision_number":3},{"id":"53ca31fe6e8f414c4cbd0bdcc33b6cf8432c4503","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 07:01:59.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9a39bda816e64e43741bca75be723f4f505d0ea9 .","accounts_in_message":[],"_revision_number":2},{"id":"56be2eabf70f1c10395d32253847fae5151fcd37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-17 07:13:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2df742bd619622b692f2ae9aadaec754cf8c802d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-17 07:19:14.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2e6dfdb36a474f488e473e93ea0e0412 : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7ba219ca94154d7e8ec8078ccc36fea4 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":4},{"id":"c71d71ab9da592823d50d841f8c50a81766ceaed","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-17 07:43:19.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/0e68a1e34bb34895ac20b32b31f837d8 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4c43ebe8903ea1d164b471c75419b0d1f2f2f62d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 07:51:32.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9a39bda816e64e43741bca75be723f4f505d0ea9 .","accounts_in_message":[],"_revision_number":4},{"id":"d89fe557fcd342b0da510d86b464374d9cfaf456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 08:00:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9c486e06c0747d3ae44a1288a7dcaa1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e3cb4c7b21043e285ad9a941ab5fb78 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f696d539c0754e92996782a0586e87a1 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb3b8720241d4317b484147c2b55dfdf : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bee5bd015771cf57b588c5fc5353b2f3fdaad0f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 08:08:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3501/.","accounts_in_message":[],"_revision_number":1},{"id":"7f9c716555c4f714773358a5b0a24160f9358ab0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 08:15:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3328/.","accounts_in_message":[],"_revision_number":1},{"id":"e922ba6ecb9704af0eb9a763458bcec9636e968a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-17 08:20:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/b6f11a1fde1d46838226a9006d1b7655 : FAILURE in 32m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ae5e160b13fbbea8ee54276aad2864cd98a7a9ac","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-17 08:37:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/980848 : FAILURE in 38m 12s","accounts_in_message":[],"_revision_number":4},{"id":"6d82cc29f9f1134a17480a55038e11abb2c674b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-17 08:42:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4d3e098199f54290a41189dfd1d55d28 : FAILURE in 38m 51s","accounts_in_message":[],"_revision_number":4},{"id":"03e28a3657b7082d540d340953cf284b5066cdb0","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-03-17 08:46:30.000000000","message":"Patch Set 4:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/980848 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"0cea6480b045cdaae4f8209dc05f66001888458e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-17 09:23:52.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-980848-3/FC/ : UNSTABLE in 5m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-48-980848-3/iSCSI/ : FAILURE in 3h 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"debda33ff9a4db9ca386c1815d135a05a7de8091","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-17 10:27:44.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1140/index.html : SUCCESS in 102m","accounts_in_message":[],"_revision_number":4},{"id":"b55f3a21149c14979805f7c1a20d1a2620ca9de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 10:29:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3f3c749af7748c199d9cb0bbb3f4ae6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fa179b97da142898df69bb0db7cb0ff : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/091b561d85c045e3bcf18daae62b4793 : SUCCESS in 10m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/977fa4cf6ced4215b363bbedd3beb152 : SUCCESS in 8m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e8831fcf2ea4079b3a891c07b31f432 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a50742828704c01a22e07cd89fc9d3c : SUCCESS in 16m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c7a1e1416e734408a48678b764169716 : SUCCESS in 30m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b1484e683db4959b58f725f15c96dce : SUCCESS in 1h 04m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4f1dd34c2b442ec85ff2a69181b8109 : SUCCESS in 1h 09m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbec4aab6f664b8eb4ed2f04aa6d9ea8 : SUCCESS in 1h 09m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fdc473c3e6e24c36a586801850a5514b : FAILURE in 24m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/21329594dbbd4954a22a8f46b94f367a : SUCCESS in 2m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a4d2c821c72493e856b45961ec68c93 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20c5ca2bb0ec43f3adc769e464a38252 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0e664bf51208480eb7f39bfaa93787cf : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d704e6351bd440a9abd9bc4dd40fb676 : FAILURE in 40m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83ce9d8627a04897a0c0942654317f4e : FAILURE in 5m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/217302f57ac745188dee78e7bd5af1ce : SUCCESS in 1h 47m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6f1a3e30259645d1881e1f19287759bc : FAILURE in 1h 19m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fc522e8a65e4c7688313f30f2757f4d : SUCCESS in 1h 51m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2b34ade0012b485d805e194c22f906a0 : SUCCESS in 1h 48m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ce391371152c49869532c1bb3479714c : SUCCESS in 37m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8996c1bfb26b488eabc7a6693b96662a : SUCCESS in 39m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb22921715b74bc4b3a63b5b62f2d219 : SUCCESS in 1h 34m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f1a5ed7f70204c41a8bdfec135331cb2 : SUCCESS in 29m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9218754fe9064c0e8f52f108a58c9fc4 : SUCCESS in 2h 11m 08s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3a77b6035ee04925b2a37be6ca1a8586 : TIMED_OUT in 3h 10m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/167da7f7f5a243bc92149d84626be223 : SUCCESS in 1h 42m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eab92496c0884ffe8994d8a2c3794249 : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":4},{"id":"2f8a8f9ff81be166f1f7204fbaa02b991d8f8307","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 10:36:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3503/.","accounts_in_message":[],"_revision_number":2},{"id":"834fb9fa6a5ac1e480502973d8ce934cbfcf1832","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 10:42:14.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3330/.","accounts_in_message":[],"_revision_number":2},{"id":"70793f9356a9ab4c79f01bd116e6dbe95615e095","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 11:55:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3331/.","accounts_in_message":[],"_revision_number":3},{"id":"bed0c265391ca54fc3aaa1dcecbb7fda75b66689","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 11:57:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3504/.","accounts_in_message":[],"_revision_number":3},{"id":"b241bb2a1aae753529d3eb6e8b884db18f35d15c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 14:26:32.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3506/.","accounts_in_message":[],"_revision_number":4},{"id":"cbe1548ffada03409cd8a07b085f01497a7c3538","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-17 14:46:15.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3333/.","accounts_in_message":[],"_revision_number":4},{"id":"f8f77f007020a0dbc4e95d504042b5ee258da068","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-18 00:24:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b6d87d7780886583cf9edbec00ee6547a3bc0c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 00:25:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6b3f062a6a5e710ac577905ab572f784bfded457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 02:20:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd22f1e64ad4f66bd96508775f55721\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070e099ed209489a959ff693e25a28f9 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e722315a1004e198e4099a8f27388a7 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0afdaed31557427280c408f980bb4213 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39dd7b1ef47b421484f30aad91a3ac19 : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/4a41a7d859ea426082bae9deff43ee11 : SUCCESS in 33m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73eec979bb494d8aae6a7ed7229b056f : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aaf603b706004459baa0c1b9d19101a9 : SUCCESS in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/653c648996a44965b898eacee46db0cb : SUCCESS in 1h 01m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/339e36645c8b498c9970ff450d46a1a3 : SUCCESS in 1h 15m 06s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/25703edb595e44d7906cd08cb31ba866 : SUCCESS in 1h 51m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fd82db01fb74b81b149817c92a65f78 : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":4},{"id":"86b9b0387039c504cb21500e598f4c7fcbfcc68d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 02:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0bc9849b8b4d866005762714ea0f662947c1e436","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 02:23:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435a88df645e4adf81d3a4a2122f409f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a25ce3e3c44624b5ed8d3df3a81991 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f5fb3f486d7e3d7a9ba2eb4bcf7e305f2c00976d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-03-18 02:58:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1159/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"9110529574800cf272fed2a1eaf9985ab888bb29","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-18 04:08:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-48-980848-4-fc : SUCCESS in 47m 52s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-48-980848-4-iscsi : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"23446d0249c208b249f7db2b450a401d36a7210f","revisions":{"4eaf907aed00839ee01725fd484f8a3dfbdfce4a":{"kind":"REWORK","_number":1,"created":"2026-03-17 06:02:09.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/48/980848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/980848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/980848/1"}}},"commit":{"parents":[{"commit":"2b20f250ccb91bfdfcf5f7414dba7e3b36157140","subject":"[Hitachi] Release notes for host group/target/wwn caching.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b20f250ccb91bfdfcf5f7414dba7e3b36157140"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:02:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:02:09.000000000","tz":0},"subject":"[Hitachi] Update docs for VSP One Block High End.","message":"[Hitachi] Update docs for VSP One Block High End.\n\nThis patch adds documentation that mentions support for\nthe new VSP One Block High End storage array.\n\nChange-Id: I9a39bda816e64e43741bca75be723f4f505d0ea9\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4eaf907aed00839ee01725fd484f8a3dfbdfce4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4eaf907aed00839ee01725fd484f8a3dfbdfce4a"}]},"branch":"refs/heads/master"},"d47e5356d653cfe8c34841a8f552d59e5fb218b5":{"kind":"REWORK","_number":2,"created":"2026-03-17 06:22:53.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/48/980848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/980848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/980848/2"}}},"commit":{"parents":[{"commit":"2b20f250ccb91bfdfcf5f7414dba7e3b36157140","subject":"[Hitachi] Release notes for host group/target/wwn caching.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b20f250ccb91bfdfcf5f7414dba7e3b36157140"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:02:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:19:19.000000000","tz":0},"subject":"[Hitachi] Update docs for VSP One Block High End.","message":"[Hitachi] Update docs for VSP One Block High End.\n\nThis patch adds documentation that mentions support for\nthe new VSP One Block High End storage array.\n\nChange-Id: I9a39bda816e64e43741bca75be723f4f505d0ea9\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d47e5356d653cfe8c34841a8f552d59e5fb218b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d47e5356d653cfe8c34841a8f552d59e5fb218b5"}]},"branch":"refs/heads/master"},"926ea85ef2f99c79b8317e14747bf8d590b57085":{"kind":"NO_CHANGE","_number":3,"created":"2026-03-17 06:23:09.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/48/980848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/980848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/980848/3"}}},"commit":{"parents":[{"commit":"2e26be2625fe13b4059ffcb6371954237ce79704","subject":"[Hitachi] Release notes for performance patch.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e26be2625fe13b4059ffcb6371954237ce79704"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:02:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:23:09.000000000","tz":0},"subject":"[Hitachi] Update docs for VSP One Block High End.","message":"[Hitachi] Update docs for VSP One Block High End.\n\nThis patch adds documentation that mentions support for\nthe new VSP One Block High End storage array.\n\nChange-Id: I9a39bda816e64e43741bca75be723f4f505d0ea9\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/926ea85ef2f99c79b8317e14747bf8d590b57085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/926ea85ef2f99c79b8317e14747bf8d590b57085"}]},"branch":"refs/heads/master","description":"Rebase"},"23446d0249c208b249f7db2b450a401d36a7210f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-17 07:13:41.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/48/980848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/48/980848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/48/980848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/48/980848/4"}}},"commit":{"parents":[{"commit":"efa2e33d53c76497ee02b1d598c005fad893c746","subject":"Merge \"[Pure Storage] Fix performance stats collection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efa2e33d53c76497ee02b1d598c005fad893c746"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 06:02:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-03-17 07:13:41.000000000","tz":0},"subject":"[Hitachi] Update docs for VSP One Block High End.","message":"[Hitachi] Update docs for VSP One Block High End.\n\nThis patch adds documentation that mentions support for\nthe new VSP One Block High End storage array.\n\nChange-Id: I9a39bda816e64e43741bca75be723f4f505d0ea9\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23446d0249c208b249f7db2b450a401d36a7210f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/23446d0249c208b249f7db2b450a401d36a7210f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
