)]}'
{"id":"openstack%2Fcinder~982319","triplet_id":"openstack%2Fcinder~master~I67feb8dbd4721ae37e47f5f840a73c0770aa09fe","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-05-01 14:47:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-04 20:25:13.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-05-04 20:25:13.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2026-04-30 15:53:19.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I67feb8dbd4721ae37e47f5f840a73c0770aa09fe","subject":"Use futurist-based thread pool for volume manager","status":"MERGED","created":"2026-03-26 21:30:19.000000000","updated":"2026-06-19 20:03:25.000000000","submitted":"2026-05-04 20:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982319-eventlet-removal","meta_rev_id":"3e3d4fe594838e6c8739f684347c45f50de4b873","_number":982319,"virtual_id_number":982319,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-04 20:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-05-01 15:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-05-01 14:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-05-01 15:39:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 21:32:33.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-26 21:42:52.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-26 22:09:16.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-26 23:27:36.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-27 00:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-27 00:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 01:16:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-03-27 01:40:30.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-27 17:15:01.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-28 12:16:15.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-04-08 16:56:40.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-09 21:25:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-04-15 12:00:31.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-04-27 19:55:45.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-04-27 23:22:38.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-04-27 23:23:52.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-04-28 13:59:14.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-01 15:39:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-05-04 18:24:12.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2026-06-19 19:54:40.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"f3d63cc256b6ac6e3906fb38fa2e5d4d993b2645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-26 21:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"744f8cd3db169544028282e8abed0fd7e28a737c","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-26 21:32:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/dff3eae25b2049b4a38cac4d12e93ad9 : NODE_FAILURE Node request 200-0000679567 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d4d7b5906d52045b31af9c2493df55968c6444ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-26 21:42:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7ac344606ce24793af2c37ca3c3f96d6 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d8be5ceedb2047a1add3525ed82cbcd8 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":1},{"id":"6ea131a8df793a5601feddc9af5c196b47bf67d7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-26 22:09:16.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":1},{"id":"94248f66fb683f4932ef25b0e0044577c6dd4993","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-26 23:27:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/a897f789322840d882304c0820b02e1e : FAILURE in 37m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1fc71cd5091369f6b717571b7c337ece1987fc44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:07:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941a9ab351944d4eba2b9c37f75fdeed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3aa5a7a7024749d78751cc3b8f94548c : FAILURE in 12m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35c39b041a094af4985aee5b32cddf80 : FAILURE in 12m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d4a868fb71314535a8f13df4b1ed6a6c : FAILURE in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5801623ecc05a475e50126a62073be468266b740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 00:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcfcf0e2f20341619cc96096dc1b4bfd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9894265f48c4a0eba370b8c8cfabbe1 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22808cb304054fdaa82be5384e440f6a : FAILURE in 7m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20eff3bd73dd40ebaae51312a93a56b8 : FAILURE in 8m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1cddc61ad9894347b279fab160e5c59d : FAILURE in 15m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f8b7a690044de8b63c1b0af99f5769 : SUCCESS in 7m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd39c1050dd24100864786f8be536164 : SUCCESS in 2m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f2a5422fc14749d5a84ca2fd190cfa40 : SUCCESS in 1h 03m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/353c96b543bf454486d58796d05e75c0 : SUCCESS in 36m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52e5db1d74e24cb5a23e1182f7b81fb9 : SUCCESS in 1h 33m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb15343b07974075aff5ea2ecc736df2 : SUCCESS in 1h 04m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f2361f7b2164e299cc084f6aa0ac991 : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/90ddc4bcb38e4778a9841d20f68e4536 : FAILURE in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/285bee5613d44cb18bd49db88b152a6c : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8bb35a4ab6a4924a3bed7e7926008fe : SUCCESS in 8m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cf858ac0f0ee4aafa3f00088fa016c0f : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76a719e24c394b4b8ae40642ca429107 : FAILURE in 23m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a37994a8fe404117996c35972daa3188 : SUCCESS in 5m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13791cbe41314922a029ac30a6f28898 : FAILURE in 19m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5c343bcaf5df463ebdde236eaf28a36e : FAILURE in 2h 36m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aff5bd3e56a245a69484078b1fed38a7 : SUCCESS in 1h 48m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d7634094e7394bedb1ea0356734dbbc0 : SUCCESS in 1h 36m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4242b9e22d740c686a26509f02b4120 : SUCCESS in 37m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/90b0878013b94e6b9cc0b449820e60c2 : SUCCESS in 1h 12m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/49964ea353f845a4ace4a53efd162429 : SUCCESS in 1h 52m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ffa2640b21046fea1d5408d002f3f80 : SUCCESS in 29m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3fda2d7fb5074bd299e77e5d26f717c2 : SUCCESS in 2h 09m 03s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/14049f3b092c4367ae061799562c288c : SUCCESS in 2h 03m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f20d4595eeb84706bbfdbffde7995d01 : SUCCESS in 1h 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bf81ba579e842c1b1f6deaddbb5272a : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e1431c138585b293c13c9ec85ba6e295bd98a351","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-27 01:16:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"858b0b09db695b386bd53c7c90825ad30c86c7ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-27 01:40:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1ffa78c148cb4b1b98a41f767914d580 : FAILURE in 40m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9b47fdac0ccb368bf4de3823f40c3d5443a6eb7e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-27 17:15:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3610/.","accounts_in_message":[],"_revision_number":1},{"id":"da11e5e304062f4a20afebd12e67f3a9f66a6120","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-03-27 17:55:13.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"8c7661a738dfc09ba0084719474a43251309b40e","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-28 12:16:15.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1261/index.html : SUCCESS in 89m","accounts_in_message":[],"_revision_number":1},{"id":"0e2835b472dd8288234d5c14cf4a47db67274414","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 11:11:23.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3437/.","accounts_in_message":[],"_revision_number":1},{"id":"799f2a41d92c169ac3ee061ce779d5c3e4587a31","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-02 14:42:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67d411487044c1a5b037ada53504908602f124a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-02 18:32:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"70e05530b408a65a0181b087d94e9196ca7d579b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 18:51:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af43c48b17b949debc802c8b0e2f9644\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35c2c3a71ceb4556bbb300cc9749d8a5 : FAILURE in 12m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/96b90fca190944e786963112fa146b9c : FAILURE in 14m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ddd44d6f6a6d4206a9150f3cd8f8d3a6 : FAILURE in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"731ffc3329180fbcb2bd0eac430adc47f4ceaa3a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 18:51:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3d395766187147beb44eb553ca910d88 : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/27ed9da51d634d179d94a26c9f2d0410 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":2},{"id":"9e8ec212fc32e8487e65528e1d2531938f408eb6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 19:11:22.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":2},{"id":"60df3771678a20f6494521ff4babcc6e05fefb98","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-02 19:30:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a63aa1580f2b498c8e0e9916f51db3cd : FAILURE in 22m 01s","accounts_in_message":[],"_revision_number":2},{"id":"459c5e1ff8c133ff65640656adacb49843c73134","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-02 19:50:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 38m 14s","accounts_in_message":[],"_revision_number":2},{"id":"2d390cfcc6414b8bc232711aa8c756e298a4c1a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-02 20:53:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9355dc93f57b5d48ec3e0e5133bc48f667490c2b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-02 20:59:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8398f4694af4d288167b57ca7d5efad : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f4e4eddd934945b9898163a22cb2e679 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":3},{"id":"70d021ffbf43f2b6f221ab36a92e9633e263ae6f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 21:30:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":3},{"id":"06b2e5bfc75415a07071f8948aab587eb4382b8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 22:12:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c635e5fe7cd4afb8c37b1780992dc24\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f4c2c251cff423b89b7f67c611b5cf0 : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed28ca29139f4ddd8bf73a5160e4216a : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/49e950741cf845b7a506f39599e8d238 : SUCCESS in 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d3adf290c14d4b95a110f4cf297faa89fad327d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-02 22:50:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 43m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bf75e7c98c15f4cb02e12f51e02dcfe11755841a","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-03 00:05:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d15e401056344a10949144cbda893665 : TIMED_OUT in 3h 00m 54s","accounts_in_message":[],"_revision_number":3},{"id":"564643d717ed18345be0ace12f702af1bb14cb09","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-03 00:52:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/8a0640e1705f40a38d6b1ad030e1f8bd : TIMED_OUT in 3h 00m 47s","accounts_in_message":[],"_revision_number":3},{"id":"7c49eb5d8bcb5684262ca483e4bfb73048163051","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-03 02:15:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/6bf0b5a6b3d44a3c97ecb1d426fd1858 : TIMED_OUT in 3h 01m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1e63271a5e678bd6fd2082b1f6fef382b838869a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-03 02:42:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e4eb93b5bda4ac0b8182626f5133cf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9ecb2d5ebf447abaf02e6b153a429d : FAILURE in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c74db1664e9945d8a3512c6132bc1b22 : SUCCESS in 11m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5de99837bb94831a2c1f9b9d8499f83 : SUCCESS in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e87838729cfc495abe4a88c4cd3e95d9 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5143a8976ec460c805f444c53d8467f : SUCCESS in 13m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5de82366b4b44a85af5fb8f7d4806b31 : SUCCESS in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/b0122a30ba5548e8b55d27bb759bcac9 : SUCCESS in 1h 02m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6fb82e300e384ac1bf58e3cd43f35fef : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b35aa06a6aa04cdc842f1d7520e7b14e : SUCCESS in 1h 24m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee1be6d1905b4cdb867c738658f42561 : SUCCESS in 33m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2638d4f5073a42b6a8aa31f2fcfc8c36 : SUCCESS in 20m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f14b8fb4f0ba4459b266b69cab89ef14 : FAILURE in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b22d10a99955478d8996855648aaf256 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3458fed16824088b76baab5d9c1682a : SUCCESS in 5m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c44e88c50e3c4b8fac929c8f36f351c8 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5453ef602b764a289cf70112b9d34b40 : FAILURE in 41m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a40a4c6134744e55afed460ecfb04442 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16c2aabbe6ed49e5a363505fb9e0f635 : SUCCESS in 1h 04m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b08a3cd0b124e49a18697da7be2d597 : FAILURE in 1h 48m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7493d032ed0e4398a523fd17a2976321 : SUCCESS in 1h 49m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/27bd4fb8f2b147fb943bbbffac1dd7fc : SUCCESS in 2h 16m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/57ca66222e544456b301c42eedb0f1cb : SUCCESS in 36m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dc784174b05f44189c758d801f4056a8 : SUCCESS in 1h 14m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/974f0be2a9244196af670b49c678b029 : SUCCESS in 1h 35m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/981fb84c0145459fa48ea0638d0ef66f : SUCCESS in 22m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f457c2e0533e4afebf9d15c7711a4ba9 : SUCCESS in 2h 12m 33s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bcc36526b0ab4478b51fc90d4d10f14e : SUCCESS in 2h 14m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c44e60e512a406986a837b90be90be4 : SUCCESS in 1h 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f33a0566424c488ebf772f876a154b : SUCCESS in 1h 03m 08s\n\nWarning:\n  Comments left for invalid file cinder/rpc.py","accounts_in_message":[],"_revision_number":3},{"id":"d896d21694400fe86f96d0360b25aa2ddd7c0192","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 03:32:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3690/.","accounts_in_message":[],"_revision_number":2},{"id":"fccd5072c27fb8c3671bc837508c9d8b10c9f588","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 10:45:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3696/.","accounts_in_message":[],"_revision_number":3},{"id":"9487ba4adf5b4d117634806d55d7774e8b0d9c95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-06 10:57:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3520/.","accounts_in_message":[],"_revision_number":3},{"id":"c98c7b2d23223b0ba0ed9bb925ef53ca42ee44bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-06 13:10:21.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6f9d745c8682ea53ab8094b4a41d1a7e431da31a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-06 13:19:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/38428cfad9ec480784f38112ebcc1237 : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9551f19dfb6e4fd48bed539c592337c5 : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":4},{"id":"87103bb85d033b782fb81c3dd52c8aabea707d3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 13:30:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/586da2692f6a4c7e9e4ee666c22878cf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fae272c4083e415c9f33998ba5a79de5 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d2f763b3ce1140d78088fcdff0475dc4 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a15262577fc74d609b25eede9b3b94a6 : SUCCESS in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a225cef7b7280f3375467c4f98199051ae6c8100","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-06 13:49:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":4},{"id":"893bf3c465b6390a2d73b2a573bdc678950e0753","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-06 13:57:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/20f79db1d36c487fb57c53a4021186b9 : FAILURE in 22m 37s","accounts_in_message":[],"_revision_number":4},{"id":"2a45a6f293a91379299a49349db805ffcea5037f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-06 14:29:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 38m 20s","accounts_in_message":[],"_revision_number":4},{"id":"8b1e1e824440e7f7c4faa3b6499ae48bada7b0ff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-06 15:32:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3702/.","accounts_in_message":[],"_revision_number":4},{"id":"75b7234bec6b8e468974be97337c9ad3fed9135c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 16:25:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b1ef43d53c749c992c71f68ca762365\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d88bc20dd941faa2e7490954a4cbea : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b02be04a269b42cb89bd976c661e3a68 : SUCCESS in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d421386fcce7426ebfc2a7a9238f2808 : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/593f175870d241028c2899fd0b6056fb : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e44ef6dceb48198be0db1d639d802b : SUCCESS in 14m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b398b00d4f14c038e71d7d046ad9efa : SUCCESS in 3m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/861ff6b2504a4375914071e6803c6774 : SUCCESS in 29m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37f25cfe66884805aaf3de8de11f98ba : SUCCESS in 51m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b6aab232c6bf4202a5fef8a98ad65536 : SUCCESS in 1h 46m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ddb3c55ab8542048813898f6ecbafbe : SUCCESS in 1h 05m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bdb55d43095a49ab8a719c9c779e7f33 : TIMED_OUT in 41m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3e7350942eb748ab8fd2975f1c8df0f5 : FAILURE in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/feb2f687b73d49c7838b5d6a5728f582 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5176607fe4664a2ab6514b45a5cc9fe5 : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c7da20742b34adca9850102c4cc71bb : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6489ea41f2f243d09185049cb97eab20 : FAILURE in 41m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7b7c66bf3b94f87b7dc071c5136d3ea : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f6f66bd2bec441b5a1c44a03fd43efca : SUCCESS in 1h 42m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f90728b2f25045e6929d58f7ba0c6ffa : FAILURE in 1h 33m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80fd9d11ccb546c49b73943609902c69 : SUCCESS in 1h 53m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/87eac9454de64a4489896a2542172942 : FAILURE in 1h 55m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3160cffe68b4e7aa4e797b0066c833e : SUCCESS in 36m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ef086c7ea8f474784047913aa577cd6 : SUCCESS in 1h 14m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5738d60c15944d628d94aef80e115127 : SUCCESS in 1h 38m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a92ba024385b4a13a7ef23f824fe8ffa : SUCCESS in 50m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/10fa490d3d874063ade0d03256559a58 : SUCCESS in 1h 57m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6cad13ccfa7b4837adbce9883695e05f : TIMED_OUT in 3h 09m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd63fb29a6334afaa893977df6525f52 : SUCCESS in 1h 49m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/936d666849cc4656be46ae769e056548 : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ff6ad192679430b4bf300566f3c5f2a62b085176","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-06 16:45:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/dea058b3d1ae4c56a19024ad5b62314b : TIMED_OUT in 3h 00m 48s","accounts_in_message":[],"_revision_number":4},{"id":"e0e2ab6e109b847d900e863d7ebde6e7fa65ba4a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-06 16:52:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bc6d8e095a9b43bb8bc05f70648796e7 : TIMED_OUT in 3h 01m 38s","accounts_in_message":[],"_revision_number":4},{"id":"24c80e9da2919f38822e522769ed560666ea964b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-06 19:03:08.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3526/.","accounts_in_message":[],"_revision_number":4},{"id":"8e1a125cddd3369e175f38fae1548ebb3c076fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-08 13:24:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5bbec8e1707d2c5ffc4f21785d3de61d88fe0d4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-08 13:46:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4abbcb53a33847a1954a77ed0bd9bd1a : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/eafc0a2ff2e546a586bce064fd8f3a4f : RETRY_LIMIT in 33s","accounts_in_message":[],"_revision_number":5},{"id":"f2353a7f41ea995917e79157024ce74be4020806","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-08 14:03:37.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":5},{"id":"0ef5d7df641dd9f9180b7f0bb29d8409758d12ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 14:33:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a3acc1a948543789f666e70549f0bd9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45c0b03914644e699c3d8d5810606216 : FAILURE in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cabcd572c12141de8b4b5a6a99bec27c : FAILURE in 9m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e5ffde602a0477c95519ff73b44d724 : FAILURE in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6d6a9fd4be9e671794d475dd7fe376e4a1533e1f","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-08 15:03:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/3f0d0e3e4f94437692c10b1c413935b4 : FAILURE in 32m 12s","accounts_in_message":[],"_revision_number":5},{"id":"dd9552244d865c94927a007072b79e7e5da0d568","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-08 15:05:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/d59ad8d331694db0bf33e8c8405b67da : FAILURE in 39m 51s","accounts_in_message":[],"_revision_number":5},{"id":"d29fd88a1cdcc0f986fabe32c11419b7a6206b83","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-08 15:18:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2d9f1929da095ade914f8982c5f198b254af86d7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-08 15:32:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 41m 21s","accounts_in_message":[],"_revision_number":5},{"id":"a89ecf8d59a36e9b8276c8341e4d9435f36f786c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-08 15:39:32.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3717/.","accounts_in_message":[],"_revision_number":5},{"id":"d5dc864f2663c39fa4e471af29769abd4575d109","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-08 16:56:40.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4e05b7f63a863b8579bae91966fa18bb2f88a6b6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-08 17:15:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"60551fd4f7ed9746207be753c5915ab6ce8eb52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-08 17:16:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"aa8a1a67a382457c07a8e87020a50806d3697292","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-08 17:23:02.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f36f11cde4c244eb883e1bff83c31986 : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/734459df792c4bd293b8d853e2b8dd07 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":6},{"id":"5ac518b585b52e44952e5cbbe6f8982b490cbdd4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-08 17:24:34.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":6},{"id":"31457bfb42720280137e6a9b8f51fdab83e8d4eb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-08 17:24:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3541/.","accounts_in_message":[],"_revision_number":5},{"id":"b314d6f44221f2763269aee0bee708b66063d9b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:32:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e78a2a01b614f54995e824b97cbd56f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3df9a7777fa84c42b148fe34e9f4a33e : FAILURE in 12m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c44df4a88f1249b98e42190e9e0efb1d : FAILURE in 14m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4ac9242cb8b14bfd8da481ad6e9b9a18 : FAILURE in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"63f62afcab2737402671a20af2cefdec3cb173fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-08 18:20:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b32f43055aaa2e2375d52e09d5cf5ea5ab60ba05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-08 18:32:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/71460feff90b4dceb5a074762ca83471 : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/59a645338ca4457d820a62e3a2e35043 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":7},{"id":"2ca59d792f89a5b6138aa382d8f693cd8fef6196","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-08 18:59:14.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":7},{"id":"3cff1f6e2908116f23cfa4ceb2fc83e4b79ec8eb","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-08 19:13:02.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-fc-driver build/6fd1bab5c8c34b12a6f71fe54f04cacf : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":7},{"id":"2bdcb53841f60fce210c9e1f1d6237aace353a20","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-08 19:28:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/237a3c0241044fa38a18a04885b67045 : FAILURE in 23m 52s","accounts_in_message":[],"_revision_number":7},{"id":"e1327550528db20747ad5d783a3170174d6ae5af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 19:32:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d157e06ae714902adb5b7723401490e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30bca6cbd92c439896f824eda5734aa2 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a18f1c3683942a089be57fe62a1d7d0 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/16b126b10de14761aa654e9d5e08bec0 : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"450d87b808a8ea790d820089a6a91591f113f854","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-08 20:04:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 39m 50s","accounts_in_message":[],"_revision_number":7},{"id":"f2c820a87e47eb087c22e966efbf5cb14489f6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 21:39:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc0830754a65441da4097c3f96dcbaf2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00588144835493598c1d19fdb4b3d3a : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e262ae1a6ed4d2787424fb4c5094604 : SUCCESS in 12m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6be43bd0be474fa2898ad431a9d60289 : SUCCESS in 7m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/99990f0aa76b468c85ead5953e1f8a75 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a757b87eea44028f8b6515f4a24609 : SUCCESS in 10m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dda3a0fd422649bbbf8149c92020fd8d : SUCCESS in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b30ce007074c49dabcfc30cf641a6df1 : SUCCESS in 1h 06m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c9348d9eac747f6abed78dfb370a285 : SUCCESS in 36m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1ed6c729e654fdd86fa7d8713facc8c : SUCCESS in 1h 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00e2687bdd0148e0958d77ed0d739146 : SUCCESS in 46m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac10c315e6ff47a9ab128fa75addbf33 : SUCCESS in 19m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/70edb6726cb748249dbecb13d4054f07 : FAILURE in 4m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b6ca5d44645497a9b51cd7853347cdd : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20ec8fbd4c334b75907b47d306dd6201 : SUCCESS in 5m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9f660e7f779d48a9a2e44c248722c397 : SUCCESS in 4m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6400056f38924299a99a355366764684 : FAILURE in 20m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe94ed5eecf4480ba7a2f9d8aad5f4d8 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fe10df4e0bb743f2aa82da4c7c073322 : SUCCESS in 1h 10m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/38d1f8014284481798a8d15e89d27dc3 : FAILURE in 1h 29m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5212a8f9e3c4b69a05bc8feca3ab325 : SUCCESS in 1h 55m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cb903e9218764e35b33ce61443dbe309 : SUCCESS in 1h 53m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/39f3a2e6b3214844853ea6bb9a6e724c : SUCCESS in 46m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd668a7333f2411e9daf4d2089dab030 : SUCCESS in 42m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e5584e784e2d4654addea9cb746df00e : SUCCESS in 1h 40m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0ad210318a8f4006be660ad664b81dbb : SUCCESS in 45m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ef01cb313ff549e18bdabb6a8ec6b1b1 : SUCCESS in 1h 01m 48s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/51ef1560a9ad4117be277b8a40a06300 : TIMED_OUT in 3h 11m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e894a9e35494346ad9e4e40dc6091b0 : SUCCESS in 1h 47m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2433118b8b4b4a3b854d34220131d4e4 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":7},{"id":"afaa40c843bf61d76855f0b01008e4e24e135637","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-09 01:19:50.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/569ba4997b3e456db59853d69e011098 : TIMED_OUT in 3h 01m 37s","accounts_in_message":[],"_revision_number":7},{"id":"10ffc39ce42e9f1a8b0e78da54a074d2933dca3d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 02:04:24.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3726/.","accounts_in_message":[],"_revision_number":6},{"id":"cea7c44051e16881ee9badba7ee6a77af18f6340","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 03:09:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3727/.","accounts_in_message":[],"_revision_number":7},{"id":"cfedc606c5684b761feff24c58093cff41d314e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 03:29:16.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3550/.","accounts_in_message":[],"_revision_number":6},{"id":"8ac7071748c82e5796ffd44731f818968a7ab10c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-09 04:31:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3551/.","accounts_in_message":[],"_revision_number":7},{"id":"2721737a4a1ef77a761d929204308fc0f42de8e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-09 13:43:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fc5d70df582d7a3ec42c33d271d209377162523a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-09 13:49:58.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":8},{"id":"dc0723c52e60a143005b83011b02f9957404b271","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-09 13:50:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9964f59c1c7d4d35b5fd47ef36875a72 : RETRY_LIMIT in 32s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7380bf0fdefa4325897430a0f75222eb : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":8},{"id":"2e0d4d2bc2d0d4c55a7c44b73232746aee1342ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 14:58:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9921fcfeaa5d45f0a4129de09303a916\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ef1b40b6a8b452380be04b9b0fefcdf : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c9d9ddd442ee4e219d72a00c7e980e03 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6af4ea1d690442819c21a73db684d9b8 : SUCCESS in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"403dd0f7298db310551fc05417022e1a7c51dce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:52:56.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0d2f23f30b748378776f6b22b6372a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76932f347f5d4b8494e633010f2eb2ee : FAILURE in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5de37091a1c04570896cb24c2256b6ef : SUCCESS in 11m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/727676c174904e0ca3720b87c06502b0 : SUCCESS in 12m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07c337c352804b13a35fae1db54b4379 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b07cabf6e8b4dcaa7f771cea14351de : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/278302121f3b44668f80335d0828d547 : SUCCESS in 4m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7d157b4593734e8da6ee52726a380cb7 : SUCCESS in 1h 03m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8716da3b9c94eecb11ce330b37a07c0 : FAILURE in 14m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdb4df416f9e4922af5af63eceb00866 : SUCCESS in 1h 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26d662e95e684155bead5d23a7642efa : FAILURE in 26m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/677894a2f8234bae855ede9ce885b565 : SUCCESS in 21m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3cabc9db69f14f118069b01982479944 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27de37f633d84bf0ae12091a2529c830 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c64da8f02e50473b93883c3e3914e1bc : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f78de3de07cd48acb9aeed95a5dfb46b : SUCCESS in 4m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d3dfb84be8645d993e9f210b01f24e9 : FAILURE in 41m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d722253c72314e019d2b2a1a6784789f : SUCCESS in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32768588d1df47a5b1ee1508b795270c : SUCCESS in 1h 57m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1e7ea4a34d74df89e4d38600bc2b86f : FAILURE in 2h 48m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/83aa9e3ffe1d43cba99561a99e0a2a6f : SUCCESS in 1h 58m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/04f78cd0bc7c443080a7a1c51eba2d0b : SUCCESS in 2h 03m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4e3d44408193460f80380dfe2cb9f22e : SUCCESS in 37m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdaf49716ef94eb5a8d1a0df103da68d : SUCCESS in 1h 22m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21910cafd1514bc88c85f2a92bfe4811 : SUCCESS in 1h 45m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b3d95904c0c8436bac2879b76a0f20f6 : SUCCESS in 49m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27f2b37ade2a44d0ad84860eb2be0a63 : FAILURE in 2h 04m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ec4b90c911964572848994001ca6b0d0 : TIMED_OUT in 3h 07m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba11de7f910542429a8954a52ddafa23 : SUCCESS in 1h 48m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d29810d6bb14c46bde10a39400037f1 : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":8},{"id":"a29b68f032e9c055584a8d8be72002088726018f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-09 18:55:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1ab4d6894ab506ae0e36654bded3facd31888d3f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-09 19:02:16.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1963ede30dd14e1380a4ea236e2668c2 : RETRY_LIMIT in 33s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e18f3bdb53c8480097bce118f7e29dff : RETRY_LIMIT in 35s","accounts_in_message":[],"_revision_number":8},{"id":"d9f5836a6eff7d3325f4a0de34ed12760cf6486e","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-09 19:41:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/cb4b306609834d29be667d234db4cd2c : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":8},{"id":"6c45e15c6ccf03a23c21fa40427b61e2a07b087e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 20:11:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0285236eba34436ab14621cd3978a00c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb56b1fa7cde43dc9d6dc742a2c6486c : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4fcb306a97164b8e91f4a3fa6d893a4e : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cc4e097eaf294f8b8dad213faa7c89cb : SUCCESS in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d684c6ff4532552f8a40ce1fafd97331d5def1e2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-09 20:32:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a0c7693c6b9746b998ab79b31356a877 : FAILURE in 38m 50s","accounts_in_message":[],"_revision_number":8},{"id":"0ea118668eeb617ce20a0b7ba1249ce716e05511","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-09 21:25:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 37m 31s","accounts_in_message":[],"_revision_number":8},{"id":"24c75c1fd10bc50ab9157f391902339f5e988ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:25:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96bc3aec1448407089394e9018355211\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42cdae99b76444dad8bdc180bf00734 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/217af1428e7a4c1e9cd140bc5980f835 : SUCCESS in 12m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a45de3a02200461ab02285af671572d1 : SUCCESS in 11m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2e6b0723afd4ed69fc8aef9ecf1fd54 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e995fe9d014245799d509a2f00259283 : SUCCESS in 7m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/422e1f24f0584c31b76039c5c5840c31 : SUCCESS in 2m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/7c1423767c634fbaac8f0de3e0ce9f2a : SUCCESS in 1h 12m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d40de746b4714fd6b19ec4d86083b097 : SUCCESS in 31m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae64366b1e974c909422f1e6e7a194a3 : SUCCESS in 1h 35m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03be996493334b6fa04f47e1cd350f70 : SUCCESS in 1h 03m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/68c241206a1644dda548ddc974e5e7a8 : SUCCESS in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/011097fbf42b415fb64b95c06ef0a021 : SUCCESS in 3m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d951d79f841c48edb228982728d96f58 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e321546dc2894d45af37d3c427a1fb90 : SUCCESS in 8m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0d70b8a924d64352a75c91aa6ada442c : SUCCESS in 6m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a0c9bbe3a1e54c40a9933d20911fa2cf : FAILURE in 41m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59b90fcb352e4c9dabb8932497a72b81 : SUCCESS in 4m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b7cf42576244b298fc2e5bfec4f5a3b : SUCCESS in 59m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96d247562c7b4a3487517562c26f9b56 : FAILURE in 1h 35m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9ec9482e2474ea3b78d85cb64f41d8d : SUCCESS in 2h 00m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/38e90a163adb4a02a2d406ae9a1d4f38 : SUCCESS in 2h 22m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/80a2d399af484c72856d651352774225 : SUCCESS in 35m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38cdaa1e5f2e4755871a5a40e9773966 : SUCCESS in 1h 23m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0018fe68e2c34001b00665773985664e : SUCCESS in 1h 34m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e1d2ec8db494059901246b9bbc469e5 : SUCCESS in 47m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/db7b6451e3cd4ce1be799eee8a8f9321 : SUCCESS in 2h 08m 01s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/f6220a5bd86a41169a60c85c43ef33b2 : SUCCESS in 2h 12m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b94fe473a044ce69f04ffaf1fca2a09 : SUCCESS in 1h 56m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f73139f262854252bc4654d9edddf612 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":8},{"id":"063092d56e3d11762e910ec347911190e22f5735","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-09 21:34:54.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_04_2026_21_28_45\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"bfc12d9175347986f42f96752ab5ca0fe58f0a94","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-10 01:56:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/772f3ecb5f4647aeae5da3d38150e893 : TIMED_OUT in 3h 01m 38s","accounts_in_message":[],"_revision_number":8},{"id":"eef0cc34603d362cb6cc7dc48263a90cb4c69076","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-10 03:46:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3569/.","accounts_in_message":[],"_revision_number":8},{"id":"3f2ede6f1b1b4092ad0b71379536718a31633a9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-10 04:51:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3745/.","accounts_in_message":[],"_revision_number":8},{"id":"a3f788cba7ef9785a6eda21a8271af4aed75acb8","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-04-15 12:00:31.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ce2a928e84217556d2ff183e65828337a4303501","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-15 12:29:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1cb1de7a0f8aabebd6892c85ef8819cca23e716a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-15 12:29:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"9d915c38f1a989c39af2fd3e4f6cbd665995dadb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-15 12:36:37.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/51b5892077e84612bebd7e56a239a53c : RETRY_LIMIT in 34s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f924d4e77ef04c3da7b93cfeba7c849a : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":9},{"id":"908822c834e261ba075b2319d51df0add180a39a","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-15 13:02:56.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-fc-driver build/2654244a9ae343fbbd940002299a811a : FAILURE in 32m 58s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":9},{"id":"40f7d4c56b4c365a1715b8958fdf37e742ede3c7","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-15 13:08:51.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/acd296eb8b4948d78757fc449e876efd : FAILURE in 38m 54s","accounts_in_message":[],"_revision_number":9},{"id":"b4e1674a92227b5efdfb98a5ba3aabccd5e4a594","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-15 13:10:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 38m 28s","accounts_in_message":[],"_revision_number":9},{"id":"448960b8a578612ab6c2c13b9f42eb479249053f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-15 13:10:53.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":9},{"id":"8886ff63abcccde6201a8542de7593685f152efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:26:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f211c6714a4e08881c7c96ec74c6d6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fdc4857603e426696d36d971ec89336 : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53f575a6c56d41348f1c31ef4198e1aa : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e6db8b10f604405b9bb1bb244bad2861 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"985c93a999d5ad5706fc488ac9064ed3d42f2a36","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-15 15:32:33.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/96ecb15bee6b49f99bb47656f9dda5f1 : TIMED_OUT in 3h 01m 37s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":9},{"id":"fae98186d1d38b047d05b9b55529c453dc2dc39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:42:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c859ce3fe224849820ea9b54b4579ad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb82527819849b29dcc3d92b461f9e0 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17198b3b0ed6450693a640ea7956d36b : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2181ecc977c347a3997a4e6ee48f2e55 : SUCCESS in 7m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6982c61265af49edb16f3770b72c9f0e : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7538b4201a447fb636aefe66cf4527 : SUCCESS in 14m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a48ddd44cfff4119b4b0aa02baf134ea : SUCCESS in 3m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/acb18b79c0c948c7a79a56478ac52180 : SUCCESS in 59m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abb4057ffd5744b1847aa49dfc5e6c64 : SUCCESS in 1h 12m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d39f58276234c46be07700d4e03f278 : SUCCESS in 1h 25m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/755fa104010148e19105b9ec4cc1458a : SUCCESS in 1h 03m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4ec2cdb21d254d8ea413983625343163 : SUCCESS in 17m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aa0c4be2d56548ad9f01316071648c57 : SUCCESS in 5m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2851494daf3e40349403e4d58a73fd02 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e33c68a00af4afeb2027bc3436a358e : SUCCESS in 5m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3afdba164b7a4ad09cda6fdbccd763c3 : SUCCESS in 5m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0685e1e92e95419185916b852b8be28e : FAILURE in 50m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/682b801edff8491aa204c6d52e50ca1a : SUCCESS in 3m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b81387716d34c3689dc8a34c1bed8ba : SUCCESS in 1h 59m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e965b74e3d824e9c8a1561a4bc21341f : TIMED_OUT in 3h 06m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5082c64109994e68b0e4c74a42e85d52 : SUCCESS in 2h 02m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/037b511964174a7dbdd0a41b55cc0c0b : SUCCESS in 1h 54m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/355c85686886422e9fcadfdcc0084549 : SUCCESS in 39m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae8a765503974f198b91a79c1385e563 : SUCCESS in 40m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fb24c8c6cd2847a0a4d530f5c5819a11 : SUCCESS in 1h 24m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32b8dcb7a72447e2a7f6afda9fc3e785 : SUCCESS in 42m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d01f14a1563140ddbf7cbe67e58d258c : SUCCESS in 2h 08m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf307f7f47044fed894cb571ae357400 : SUCCESS in 1h 26m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d1e040f2a724ce2b3dd0cd29f5cb650 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":9},{"id":"6f1ac36346e5cc0968315492133d303d3246e6e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-15 15:52:19.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_04_2026_15_46_10\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"a5c86a6f3ff24a6ddc472ce88bd3730b06399f46","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-15 16:18:53.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3644/.","accounts_in_message":[],"_revision_number":9},{"id":"815adb60ba478ebba6aab0759f003a37d62f88cc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 11:44:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3820/.","accounts_in_message":[],"_revision_number":9},{"id":"a3aff317e700af34bfe599c14ac684a6205cd12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-04-16 14:37:30.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"a013f03c6606f761c6939495dd6061fe0ed1a024","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-16 14:41:25.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/de92fd56d48749b5a65a8579ac6d66b2 : NODE_FAILURE Node request 200-0000707331 failed in 0s","accounts_in_message":[],"_revision_number":10},{"id":"b652c7f44036bb37e628da9cb89c418aa852ea89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2026-04-16 14:46:02.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"fc11faef245692e64712a14789dfa47d17e82697","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-16 14:49:17.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a05494fef7e1469caf72d4b7e04a610a : NODE_FAILURE Node request 200-0000707339 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"2c4b51479e4328c07485feef77232683212b5fbb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-16 15:12:59.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/cdc3a999971947f68baeb210911b4f84 : RETRY_LIMIT in 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/18e35527621a4535b22a84b6c3e0e579 : RETRY_LIMIT in 34s","accounts_in_message":[],"_revision_number":11},{"id":"6b1a0f1acfd742c93896112b50a9bd5e903cc67a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 15:46:24.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":10},{"id":"9b5506f85b901fe37eb937e9d8698212d8d9ec8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:04:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d9f393599d46a8802c7e6a5c38ebdd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01eaf1f71d8a4ef78613047c3788cf29 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c7dd835f7254cbaa16f232600b5bbb3 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/07b003702bde403c94eaf7bdae99bed2 : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e87658bfcd41da3440eb4f2c0add9e9604aacc0d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-16 16:58:53.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":11},{"id":"3cd4ee5ab2dd95c3296cd0663b07af101d402027","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-16 17:54:29.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/b1cc1a2f0fe847b4a896c87e5adb5720 : FAILURE in 52m 21s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":11},{"id":"07be30540e2a0654933f3b448de24d8098e9a9d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-16 17:59:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3661/.","accounts_in_message":[],"_revision_number":10},{"id":"1e0181300b9ca02fddb8baf40b74537932e2b534","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-16 20:45:55.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : FAILURE in 42m 20s","accounts_in_message":[],"_revision_number":11},{"id":"7a5b95b48fe4025a99da97d52554a106d999744f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 21:43:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3b98f33a0a4e99ab3481a064e748e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b145abeb5bfb4b87843d7d198fdd06b8 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd701bfe609a4ae9ac66d5f86b20bdf2 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c6449b5bd184fc288ced1fbea44ef3d : SUCCESS in 11m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82110ac96f114c65947c20e6818fa305 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415954a6e76b4507af56c37e67981244 : SUCCESS in 13m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7a749d24c364ab5a05bbab84aee61fe : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9411061fe89e4184bafa8b5526dcf934 : SUCCESS in 1h 09m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/586a8333b1814453997a26c3ec924d33 : SUCCESS in 34m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb80580819b14d3992bffbdbdf99537c : SUCCESS in 1h 44m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa9a44dce12a4bc99959ebf7baeeb772 : SUCCESS in 1h 08m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9370613d4f142e09190e914b497fd8c : SUCCESS in 29m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d94d57bbd0b24c52ad6d245757def822 : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4379abc905eb4ae5875d66ef9bba1c4c : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e3ebcd02a084726bd9e4699525351f7 : SUCCESS in 5m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eae9d3ec65334aadabd33d4861f58d29 : SUCCESS in 5m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/06120185ac27454aae8a4934aa94ebd7 : SUCCESS in 41m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/24c9c53d2b324c4c8e398f3f9bc93a66 : SUCCESS in 3m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67e42d72053f4c8bbcffaabe81a96c3f : SUCCESS in 1h 20m 48s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e9e2e7a8e36a4dfe88be1a08a229c358 : FAILURE in 2h 56m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b93f2aa7450343e7bdc5cb8843492d13 : SUCCESS in 1h 51m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55a5315144c241a6bd3a2c3ecc39f5c1 : SUCCESS in 2h 06m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a2722de5c0d84d018da1a2d867e85442 : SUCCESS in 22m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c899fa06dd04d1e9c50699707093af6 : SUCCESS in 1h 07m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ff6d9b96d9d846138e6ebd5cf3433a62 : SUCCESS in 1h 52m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa44fce58c6e4f0d878832b579b81d0a : SUCCESS in 51m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4674102b1d264140aaf80c3f3f71f29e : SUCCESS in 2h 00m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc8535ea6acf48cdba669d36611e9a1b : SUCCESS in 1h 43m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c78b3fa659584134b3304b4143a9415d : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":11},{"id":"51f2a2cf051f83b4199cb270cea0fabda3bb576e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-16 21:53:56.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_04_2026_21_46_56\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"9c730193be7839119b7c81d6028f05800a3ac1f7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-16 22:09:00.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/71ba941623464043abdb5cb1ccaed04f : TIMED_OUT in 3h 01m 39s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":11},{"id":"6a9b275e8d2d48bd2c0bd851403997695a808dc4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 04:14:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3669/.","accounts_in_message":[],"_revision_number":11},{"id":"26a316a4a574fe6c85700d1f587ece262f75d4f6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 07:11:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3837/.","accounts_in_message":[],"_revision_number":10},{"id":"14902d276025a06a95921f04fc7fbf5478579568","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-17 17:05:04.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3845/.","accounts_in_message":[],"_revision_number":11},{"id":"110b10cf92cd3c6c5caad744aba615f85294784d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-27 19:46:01.000000000","message":"Patch Set 12: Patch Set 11 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":12},{"id":"3f1ff9f161ca8a72e83433be92ce1f96cf6f9e4a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-27 19:55:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-fc-driver build/d81b43a312b8462eb7b8c6e45672a4a4 : RETRY_LIMIT in 15s","accounts_in_message":[],"_revision_number":12},{"id":"fa494928e1aea1087c7fb000b1ecd8d4bcb0e85f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-27 20:27:13.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":12},{"id":"7303675d4ad42f4ba72cc8ab227984e89a9ee046","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-27 20:46:04.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/cad4b032f52c40b49d0001ab29dd1e48 : FAILURE in 23m 24s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"404df515232f83b81e60562e7caec2a58f83ef4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 21:01:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63f6c62eba804ad9bda58e39048a5d03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72998c1c02e54e0fb19489ed85972ae0 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfb0575743a24554ad9f5f95ce975541 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b777ef958808422a9670a1ca31cb9a96 : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"de29686bcc31370138cdcc4ee9bea3955bf19fd7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-27 22:05:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3978/.","accounts_in_message":[],"_revision_number":12},{"id":"c7df573082dddf3e6cfc4504577fffc4c0e10f0f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-27 22:17:47.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3802/.","accounts_in_message":[],"_revision_number":12},{"id":"03d6afea9c0d56c5d466a81330e13e7b6870eb66","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-27 23:22:38.000000000","message":"Patch Set 12:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"cedecefb1ab9f260b7e479972a31e6fd7cd8b439","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-27 23:23:13.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982319 : TIMED_OUT in 3h 00m 44s","accounts_in_message":[],"_revision_number":12},{"id":"ddd46283868524f6653f32339c47496f6d30d283","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-27 23:23:52.000000000","message":"Patch Set 12:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"adacb3b8f6e480aba9794ec6e705236b06f07f24","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-27 23:29:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/37efef43d4634c70b54fce0f37731901 : TIMED_OUT in 3h 01m 36s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":12},{"id":"516bf36207c40b1c8f67fc686980910050c4abf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 07:13:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e7abcf3ea364ee7bfbb03517fdd20a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796f703e7de0494aa36bcd092017d35b : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0b55d76c6b7489ebb464cec8c03f3c2 : SUCCESS in 10m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2878903b60814f809560d7e6c9c30ab5 : SUCCESS in 10m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dea173a87d584a8ca52e511c18fbd432 : FAILURE in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6787d2732fb45ee841f54523d551347 : SUCCESS in 14m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a99eca2fb33443aabffdb07b3f51ff8 : SUCCESS in 2m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/b791fd2253c9403e941f5d3d624e722a : SUCCESS in 30m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21dc24f4bb1d46df9d5ec08e3c13a7c8 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f96e5fc1ffa4db5b8c579c66ca41157 : SUCCESS in 1h 27m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d69aad68af3f49c49f3864de5c9f89c4 : SUCCESS in 40m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f39b297eb176404ea842863734e5301d : FAILURE in 25m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bc29c056c9ce484d927905f8087a402b : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc522258843c4eb68a8aa3a18df5a9dd : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10e2aad97675483c8fbb38bf8e0a1d5e : SUCCESS in 5m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/df612090ebf64c219935c94c8a485676 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fc79b7068a7647e4b7d596c9d521eca2 : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcfc2655ddfe416b8eb2df7700201d22 : SUCCESS in 3m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7c1f472479bf4900bc3f10b19d1602a0 : SUCCESS in 1h 38m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0dfca2f962814a2880e504c2b9d02c4f : RETRY_LIMIT Host unreachable in 3h 41m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c512d0f106de497cb6948dbf119fbd6a : SUCCESS in 1h 43m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1d67e7d0bc2d445dacfbe945d0a631b9 : SUCCESS in 2h 09m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d48f83a0135444bdb09f51963679023c : SUCCESS in 36m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86bcf3cac2074b929cd09ea72a267317 : SUCCESS in 54m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9e305bc039b4cf0be7e224cd9eb55df : SUCCESS in 1h 40m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6f947f04a7c24e42b16931e211c1b3b2 : SUCCESS in 31m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e36914e812264d1392367e1e3f58374b : TIMED_OUT in 3h 07m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa3c12fbb096424a9339274fb5a0f805 : SUCCESS in 1h 28m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edb7ed2c35db4ce99fb3d66dd59755c9 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":12},{"id":"9d1736a583eb75081bb7fa9a9bc3c663d35af130","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-28 12:54:16.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"01d9808d0ca68b5866a435f8b066ab12e6066104","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-04-28 13:59:14.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-19-982319-12-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-19-982319-12-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"7bda1b1748be1deec815eb67cf82b36a518a9072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 14:47:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305a6564ab824d09a1e5b925a5fce14a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c3c6f8c430a40ceb1a06a7c948da65a : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c92170d250a4fd5b6554ae7488a192f : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/91eae2529800457f9d9ca18e59f2b5aa : FAILURE in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d62189f4fe3a5f3f736b427de06c8455f01e35ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 21:53:03.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd1c8ae2c1b4daabb2f11d8f947c29d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ba8ea970d9492f8991d8336ffbf587 : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfb95df744fc40319980a1b1f276e64a : SUCCESS in 13m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c25ca9cf23704bc6862a966e1dd5c328 : SUCCESS in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a47751525ec04c6990f1b03c746e021d : FAILURE in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903a8fd1b28c4b4eb7c442932fcdba0d : SUCCESS in 15m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8caf8482d1ac43b6b313b6553c1b33f6 : SUCCESS in 3m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/6d85eb6e5fa34b5dae1bb299b6dbe82a : SUCCESS in 1h 06m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2519a1879a5b4716b7973d47f9ef5f52 : SUCCESS in 1h 07m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a60cad4686e421f9da45ae17153cfaf : SUCCESS in 1h 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55e7dc12382e4e73ad2cb5f251ec91ef : SUCCESS in 1h 05m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/42ee5b3a270b418793724fcea68523f1 : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8cde17c55c244ee88bf5160fa89926ba : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e4ceebcb2424527897637cdc95a27b7 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4173fdca01664879a35ce9f55499af31 : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d19eb7137c044af3928ab1f2f941c838 : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0393bb97baf8429db1cfbf893b3ca395 : SUCCESS in 39m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea29885c9a4949e98827b1d4b6830bf8 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/26aadc02c24549ef8755fe3d5ceead79 : SUCCESS in 2h 03m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a816a776f060463ba4733eed5ef72955 : FAILURE in 1h 46m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00a7ed0f837a4a0a9f29d9bd6e84d84e : SUCCESS in 1h 49m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d3458e1abdfc428eb09620b911075409 : SUCCESS in 2h 06m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/17473328636545cd835ca01a96408aa6 : SUCCESS in 37m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c0a3b900be524cd0ab0e8cbe24eabd9e : SUCCESS in 1h 15m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/125bab45950c429f887721680beae3a9 : SUCCESS in 1h 48m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ed4b3059e09346e8b180d1753bfa08d6 : SUCCESS in 44m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/438fd3a5aa7d44fe9a2da86229cb546c : SUCCESS in 1h 55m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a7484f4829e4af1b081822ba2d79bb5 : SUCCESS in 2h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9aa9237798464ab19b4431f7b4a7f4 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":12},{"id":"444caea71da33df9f663756cc5fbc8f765ec34e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-28 22:02:29.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_04_2026_21_56_24\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"c254f79f465a78fadb9374a4200aada7052b7de3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-29 16:36:53.000000000","message":"Patch Set 13: Patch Set 12 was rebased on behalf of \u003cGERRIT_ACCOUNT_30555\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"}],"_revision_number":13},{"id":"5b07607fd24c7e5ca45cb866a3b9de1f6afb6308","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-29 17:24:41.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67feb8dbd4721ae37e47f5f840a73c0770aa09fe .","accounts_in_message":[],"_revision_number":13},{"id":"f2754c678e0fbadf533e8e2fd6b24185dae4d993","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-29 17:45:34.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hpe-fc-driver build/98baf93145c041b6bce4b779bac99353 : FAILURE in 32m 05s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":13},{"id":"cbbe390f5cd432b390157fec00da469cfa65e339","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-29 20:33:36.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"8b3828090e9a9a53832c426f4a0f0a992da902de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:49:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ecaf05031284f25b60f7365f30568a7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fb7baa37db844559386acacc7acc7dd : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/06d214309baf41e4b5d38f123c3fb97a : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/82d03c79ec5d49308d25dc87d6d1c503 : FAILURE in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cac6afe9840fff9b0e753bf600bcf972471798ec","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-29 22:54:34.000000000","message":"Patch Set 13:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"7147b269bbc65dbc3b2a45228fd58c5843915922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 00:37:49.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53f1f52dc48f43b6b368883a7801167d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/966c660e7d114e51a3390d9f63412735 : SUCCESS in 7m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e14cb904e73d4dd3908ad092aa1bf087 : SUCCESS in 10m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5da94892053d4759b548cf65606c61b2 : SUCCESS in 10m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48d228b8181d4055aafb011b9851c79c : FAILURE in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a3b542a64a4f61add5074672a8d628 : SUCCESS in 14m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/037e8b73f84b441dad9bf8543bc5cd40 : SUCCESS in 2m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3bd7a1c8db284e1eba9346135146bccc : SUCCESS in 1h 07m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bef8f5a998df491aa028d967509a9507 : SUCCESS in 35m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87979cbe231a41bfb74841a45e29fd54 : SUCCESS in 1h 15m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bacf845aba684c7eaa07c2cabb5edef5 : FAILURE in 1h 20m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ce2b37517264cf99a2a8e42f7a806a7 : SUCCESS in 17m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/416a854b73d74e999943a3d5bdbef3ee : SUCCESS in 3m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8bf89c10b211494f9709a3993c9222fc : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4c976668ddb4d93a22e3ed33c22d92e : SUCCESS in 7m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4172f7ea97944c8db7ea17dd7d04aef6 : SUCCESS in 5m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be4016828bf04e4e88220d2f88151f52 : SUCCESS in 48m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db3b30c5ee5247689f8b12e6bd103034 : FAILURE in 4m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/380985784b4c4ada8017ee0e6c11a753 : FAILURE in 1h 16m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eba453e6a2db44cbb78893f4c2093eb7 : FAILURE in 2h 39m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11dbad8d1e66438f905d91e2f4a54f33 : SUCCESS in 2h 16m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/12c2ddd305ce4a7ab043fa5b3340de15 : SUCCESS in 1h 43m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fd419adf4f9347cd91099b0660154434 : SUCCESS in 36m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63a6a7500ea24a37ba709bada05db6a7 : SUCCESS in 1h 05m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/29cc485b14fa4aaa90a185661d3ea355 : SUCCESS in 1h 31m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3ba959544e14394acd4d8b72de3aa67 : SUCCESS in 34m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9cb2b2e49d944f208440a502c8ee576a : SUCCESS in 1h 02m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8320d337ef1e4fbda48843fadfdedd1a : SUCCESS in 1h 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f8ac46d555443318dc608d375d06242 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":13},{"id":"25de5ca573b45c9948e677d21f88918990ebd84c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 01:28:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3838/.","accounts_in_message":[],"_revision_number":13},{"id":"5f1a5059c01cab70fdb550e69deea61f674a2b05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 03:51:01.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4014/.","accounts_in_message":[],"_revision_number":13},{"id":"52a2013110e26a059cc70886cca762a338282a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-30 14:49:19.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"1b6b44cbd3e0530830619762172e36b14c08cb14","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-04-30 15:53:19.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hpe-fc-driver build/7407e9e476cd4057849f01fce55789eb : FAILURE in 30m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":14},{"id":"24f7d63ce5cccd4fb8a37aedd70f71e570ed14d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:31:32.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff757908e3f547d78f6edea7f55071fd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/202dd82b4932422e80686b698c36bfa4 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/28e045dd8a8f4a0ebae2dfbe76d6c9e6 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/29664046d82b4b8e9a96fef2a5fc8682 : SUCCESS in 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b74e9e0e6379155e7ef646cde532175f35d86e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:13:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5248dbf382a49cf8f46f5446e66ebcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7497e47cd1dd4153b0fc7681bfa31c45 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/821074c8fbe74612a710ab5621f1b63b : SUCCESS in 10m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce097702f5344c2f8bfc1f3b6ed4f304 : SUCCESS in 11m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b4b60282c6d4d32a714fc02fb79a6fb : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc05f87420bf45fab1942c68f7e6f467 : SUCCESS in 13m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/301591fb84d7424eb5ddf120547690bf : SUCCESS in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/098e65b4fa684d25a42569f78bb3277b : SUCCESS in 53m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0adf88a94f2240cfb513840e390fc69e : SUCCESS in 44m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/847da647a4a2496cb305154928a1799d : SUCCESS in 1h 43m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11240a31bdc34cceabb1722820d5222e : SUCCESS in 44m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/94634b5e9cfa4526b9769ced16e1785b : TIMED_OUT in 41m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/664c1c23d4e647f1969972022f98c72e : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d88d0ec24df4b4d9f65600d13e83c07 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d36c12aee0d459a80e90cc447144479 : SUCCESS in 7m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14a1994b5cc54973b9c3caa6562fbf2c : SUCCESS in 6m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a591533e18624ae896d2ff8b29301769 : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/892d0d9dfe134d64a24a9633aefd1c92 : FAILURE in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/155ea67a65c044efbf35d8f8caa43b47 : SUCCESS in 1h 46m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c61076552e4e4e86a09e2c548805e918 : FAILURE in 2h 14m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2558dff94db4e75b0363f360e944c79 : SUCCESS in 1h 44m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b178da07e629486283a6fa83922d6417 : SUCCESS in 1h 46m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/179239a2c5ff4d34b560ba5ce89b444c : SUCCESS in 36m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b4c8e37370474da4a6e07f6e571312cd : SUCCESS in 1h 17m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/288d6afef2de47f4b0eeaedb0623c092 : SUCCESS in 1h 20m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/78ddadab7f8c4c1bb5885fae73fe5a84 : SUCCESS in 38m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5039f9f93aef45ec92537b5d42546de9 : SUCCESS in 1h 54m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5be33cfdc96749adbee01565f1c135ab : SUCCESS in 1h 31m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e4fe9d8fdbe4bf58f017261cacd06b2 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":14},{"id":"8dd4f4c57b4772be2bca337d7e09764370d6cdf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-30 17:23:13.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_04_2026_17_17_07\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"b0ba18744da83a3e7e27641a7709e21c133c5e38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 17:24:44.000000000","message":"Patch Set 14:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3847/.","accounts_in_message":[],"_revision_number":14},{"id":"ec2d40390cc0fc399cd554be3a278a2eaaca8b32","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-30 17:28:15.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4023/.","accounts_in_message":[],"_revision_number":14},{"id":"07a8a19e16bd58f65bd596ca4bba9228cddd645c","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-30 18:27:58.000000000","message":"Patch Set 14:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"448425c9cb8a8c56479dc90c603ccba1355d69eb","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-30 18:28:37.000000000","message":"Patch Set 14:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982319 : FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"68dbeb6c73e2e2d8103bde82d20355ea5f0a3dce","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-05-01 14:47:12.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"389c0e2c522faac85f4348c7b96e91c2600c4580","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-01 15:39:48.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8e06d4c9db2cc2300ccf4cbd752bec99d6affa18","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-05-04 18:24:12.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/970332 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"c1ec115f1b6bd48d112db210045162c51dd19e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:24:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"af01a4a1af11b966ac49285ffd2540b5768aad56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 20:25:13.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85249d2339aa4c0d88faca7ce8a42f5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027da75b4be1478094a9431ee99810ea : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebf9c648a8df467b883d0a7d7d9d3b60 : SUCCESS in 10m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9ed713253dc47b39fbd69ffe930d817 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61711ed2f0464a7c8cac91ce5b21e251 : SUCCESS in 11m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/28371978f19545f59598173fd85b6ffb : SUCCESS in 2m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/99052a7dd201431cbd52575a0cee9f65 : SUCCESS in 1h 09m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63849c8878964ed6a6d661bd73224732 : SUCCESS in 1h 03m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/041a3eebf51b44a288f5288e04e72f9a : SUCCESS in 1h 38m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3ecdfe85004427caca773eb2561371c : SUCCESS in 1h 10m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/26b391b57f194d3ea0e5ff3f6cbdac16 : SUCCESS in 1h 14m 58s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59f7d583860343c38f3abf86ad68b037 : SUCCESS in 1h 47m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f96cb4c60c94702b2b45dafef1955ec : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":14},{"id":"495c45a4a8c9ee1da0061e6ba77350d903eff6be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 20:25:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"c409d9fa49f634fbb93ceb648ebd649a912848ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 20:27:26.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9be269025546f6921990f37cacb38e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b2df47b5fc4a77ae0c2f64c0dca9a2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":14},{"id":"851725387e7c3e0e647a8138d82f2f2e365ac75b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-04 22:24:03.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-19-982319-14-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-19-982319-14-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":14},{"id":"931299cc30c93350500193e2af63dad42f37056a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 11:22:19.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a09e6c97f6d4498797aeb8d2bb4dcb27 : FAILURE in 29m 32s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982319/etc/tempest_log.log","accounts_in_message":[],"_revision_number":14},{"id":"b3c6e39735b084e47c70ed94cd20356ff19da7bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:54:40.000000000","message":"Patch Set 9:\n\nBuild Failed","accounts_in_message":[],"_revision_number":9},{"id":"42aacf0e68f1702176b54e3fffafe84f3729038f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:54:50.000000000","message":"Patch Set 7:\n\nBuild Failed","accounts_in_message":[],"_revision_number":7},{"id":"b37ed40386f226d44062fb37f4d75cfc94f3af45","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:54:57.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"84b4f8a9acf4ccc56cd08e989606516632585bdd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:54:58.000000000","message":"Patch Set 10:\n\nBuild Failed","accounts_in_message":[],"_revision_number":10},{"id":"aebb64dfd5f33f430d93b0b5418212cbab177410","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:23.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6},{"id":"9db083c747b8f0bdee44eecc2805c93ba93335ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:37.000000000","message":"Patch Set 8:\n\nBuild Failed","accounts_in_message":[],"_revision_number":8},{"id":"a886330d179dd24a30163eaa9a8751038f8fecc7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:59.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3},{"id":"e036f2803ab1102755c399a2799c7ecf6272bcdf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:10.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5},{"id":"ba0338a7f138bb7f183d25cd7ea06d39eeac2afa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:01.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"745a1f7c80cb4496d0953a071d2a9327139128f3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:57:06.000000000","message":"Patch Set 12:\n\nBuild Failed","accounts_in_message":[],"_revision_number":12},{"id":"37fdde6de4de77cc88f417b599ed861aefefea01","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:04.000000000","message":"Patch Set 11:\n\nBuild Failed","accounts_in_message":[],"_revision_number":11},{"id":"aeeffc4d4709a3289cabc94518127f78fc74e9ce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:12.000000000","message":"Patch Set 14:\n\nBuild Failed","accounts_in_message":[],"_revision_number":14},{"id":"8a4a76494bcea0e2f9b9a01c48892dbf55809d8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:17.000000000","message":"Patch Set 13:\n\nBuild Failed","accounts_in_message":[],"_revision_number":13},{"id":"3e3d4fe594838e6c8739f684347c45f50de4b873","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:25.000000000","message":"Patch Set 14:\n\nBuild Failed","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"91ea6755b432bca6d2e73b133e3b26ff60e2aea5","revisions":{"74497b3bfbe134565e630908addc0a2aebe35932":{"kind":"REWORK","_number":1,"created":"2026-03-26 21:30:19.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/1"}}},"commit":{"parents":[{"commit":"7cf05b0b7eb0317de745072e5237d61eae7daea2","subject":"Replace _add_to_threadpool with native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7cf05b0b7eb0317de745072e5237d61eae7daea2"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-26 20:58:11.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-03-26 21:29:11.000000000","tz":0},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74497b3bfbe134565e630908addc0a2aebe35932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74497b3bfbe134565e630908addc0a2aebe35932"}]},"branch":"refs/heads/master"},"2db3413ca3454807d94713008e49a74443ee01e5":{"kind":"REWORK","_number":2,"created":"2026-04-02 18:32:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/2"}}},"commit":{"parents":[{"commit":"dc721f2e0c32e0a72fbd7be9ebdfd7daa5b8c20f","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc721f2e0c32e0a72fbd7be9ebdfd7daa5b8c20f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-02 18:19:55.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2db3413ca3454807d94713008e49a74443ee01e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2db3413ca3454807d94713008e49a74443ee01e5"}]},"branch":"refs/heads/master"},"148a5c111fc32454af5638e2eec0accd49e00db4":{"kind":"REWORK","_number":3,"created":"2026-04-02 20:53:00.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/3"}}},"commit":{"parents":[{"commit":"dc721f2e0c32e0a72fbd7be9ebdfd7daa5b8c20f","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dc721f2e0c32e0a72fbd7be9ebdfd7daa5b8c20f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-02 20:52:36.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/148a5c111fc32454af5638e2eec0accd49e00db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/148a5c111fc32454af5638e2eec0accd49e00db4"}]},"branch":"refs/heads/master"},"edc1221cabc5ef6441dba19a08d2e5bd82892fa7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-06 13:10:21.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/4"}}},"commit":{"parents":[{"commit":"5cf73400fb2ec5a5d10af8a11e3952c7e35ad0fc","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5cf73400fb2ec5a5d10af8a11e3952c7e35ad0fc"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-06 13:09:39.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/edc1221cabc5ef6441dba19a08d2e5bd82892fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/edc1221cabc5ef6441dba19a08d2e5bd82892fa7"}]},"branch":"refs/heads/master"},"350254f59cf8ee4b87a806d5e16a03692099e207":{"kind":"REWORK","_number":5,"created":"2026-04-08 13:24:24.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/5"}}},"commit":{"parents":[{"commit":"cbad4a6e598cc5277d56eba99a7ff5c9166b60eb","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbad4a6e598cc5277d56eba99a7ff5c9166b60eb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-08 13:23:36.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/350254f59cf8ee4b87a806d5e16a03692099e207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/350254f59cf8ee4b87a806d5e16a03692099e207"}]},"branch":"refs/heads/master"},"201e7efe2c79b48c0379ec55d3ab5d4edeaa04bc":{"kind":"REWORK","_number":6,"created":"2026-04-08 17:16:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/6"}}},"commit":{"parents":[{"commit":"cbad4a6e598cc5277d56eba99a7ff5c9166b60eb","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbad4a6e598cc5277d56eba99a7ff5c9166b60eb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-08 17:16:00.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/201e7efe2c79b48c0379ec55d3ab5d4edeaa04bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/201e7efe2c79b48c0379ec55d3ab5d4edeaa04bc"}]},"branch":"refs/heads/master"},"eb8c7d37cf277a48d09829d0a31ed879a3adcdb6":{"kind":"REWORK","_number":7,"created":"2026-04-08 18:20:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/7"}}},"commit":{"parents":[{"commit":"cbad4a6e598cc5277d56eba99a7ff5c9166b60eb","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbad4a6e598cc5277d56eba99a7ff5c9166b60eb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-08 18:20:18.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb8c7d37cf277a48d09829d0a31ed879a3adcdb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb8c7d37cf277a48d09829d0a31ed879a3adcdb6"}]},"branch":"refs/heads/master"},"58498762edf871f52d7963156800275c668bb889":{"kind":"REWORK","_number":8,"created":"2026-04-09 13:43:54.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/8"}}},"commit":{"parents":[{"commit":"cbad4a6e598cc5277d56eba99a7ff5c9166b60eb","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbad4a6e598cc5277d56eba99a7ff5c9166b60eb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-09 13:43:21.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58498762edf871f52d7963156800275c668bb889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58498762edf871f52d7963156800275c668bb889"}]},"branch":"refs/heads/master"},"cce6d9617a079b607d927f148c9b9b0b0d76dc10":{"kind":"REWORK","_number":9,"created":"2026-04-15 12:29:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/9"}}},"commit":{"parents":[{"commit":"cbad4a6e598cc5277d56eba99a7ff5c9166b60eb","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cbad4a6e598cc5277d56eba99a7ff5c9166b60eb"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-15 12:29:02.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cce6d9617a079b607d927f148c9b9b0b0d76dc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cce6d9617a079b607d927f148c9b9b0b0d76dc10"}]},"branch":"refs/heads/master"},"22b18c556843355ac69652af7288c2ff5482a80a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-04-16 14:37:30.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/19/982319/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/10"}}},"commit":{"parents":[{"commit":"1b1d1d71fec8404e8ace9ffcf0baf27df01dd3d8","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1b1d1d71fec8404e8ace9ffcf0baf27df01dd3d8"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-16 14:35:57.000000000","tz":-180},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22b18c556843355ac69652af7288c2ff5482a80a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22b18c556843355ac69652af7288c2ff5482a80a"}]},"branch":"refs/heads/master"},"3c4393f755c03da81d65b6e26610e12dae34862a":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-04-16 14:46:02.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/19/982319/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/11"}}},"commit":{"parents":[{"commit":"470115b5071e7ff20dc4f0ac7dcbcd398d3d8bd7","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/470115b5071e7ff20dc4f0ac7dcbcd398d3d8bd7"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-16 14:43:02.000000000","tz":-180},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c4393f755c03da81d65b6e26610e12dae34862a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c4393f755c03da81d65b6e26610e12dae34862a"}]},"branch":"refs/heads/master"},"bbccde0aa9033d38f1998c2e515e48294f7a7b51":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-04-27 19:46:01.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/12"}}},"commit":{"parents":[{"commit":"4c14113b26d52d33583ed636962b763ee1cb0a35","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c14113b26d52d33583ed636962b763ee1cb0a35"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-27 19:46:01.000000000","tz":0},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbccde0aa9033d38f1998c2e515e48294f7a7b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bbccde0aa9033d38f1998c2e515e48294f7a7b51"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3c4393f755c03da81d65b6e26610e12dae34862a","theirs":"4c14113b26d52d33583ed636962b763ee1cb0a35","contains_conflicts":false}},"8f025648f632abfb24789f88e9638740046e954f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-04-29 16:36:53.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/13"}}},"commit":{"parents":[{"commit":"a8bed30b362dcded145c12ade061ae25c200f947","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8bed30b362dcded145c12ade061ae25c200f947"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Fernando Ferraz","email":"fernandoperches@gmail.com","date":"2026-04-29 16:36:53.000000000","tz":0},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f025648f632abfb24789f88e9638740046e954f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f025648f632abfb24789f88e9638740046e954f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"bbccde0aa9033d38f1998c2e515e48294f7a7b51","theirs":"a8bed30b362dcded145c12ade061ae25c200f947","contains_conflicts":false}},"91ea6755b432bca6d2e73b133e3b26ff60e2aea5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-04-30 14:49:19.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/982319/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/982319/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/982319/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/982319/14"}}},"commit":{"parents":[{"commit":"cf5a8cd05609c614bdd893d7de1bd64caa250ef9","subject":"cinder/monkey_patch.py for threading mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf5a8cd05609c614bdd893d7de1bd64caa250ef9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2025-07-15 19:48:50.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-04-30 14:47:10.000000000","tz":-240},"subject":"Use futurist-based thread pool for volume manager","message":"Use futurist-based thread pool for volume manager\n\nThis allows setting the thread pool size for\nboth eventlet and threading backends.\n\nChange-Id: I67feb8dbd4721ae37e47f5f840a73c0770aa09fe\nSigned-Off-By: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91ea6755b432bca6d2e73b133e3b26ff60e2aea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91ea6755b432bca6d2e73b133e3b26ff60e2aea5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
