)]}'
{"id":"openstack%2Fcinder~982620","triplet_id":"openstack%2Fcinder~master~I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f","project":"openstack/cinder","branch":"master","topic":"bug/2144863","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-04-01 01:34:26.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36711":{"account":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"last_update":"2026-04-01 08:18:21.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-04-01 01:34:26.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f","subject":"Fix backup service states cleanup","status":"NEW","created":"2026-03-30 08:42:05.000000000","updated":"2026-04-24 17:14:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4de18d3446060ae5a13c91879370017d8c517c2b","_number":982620,"virtual_id_number":982620,"owner":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-01 08:00:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 08:42:44.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-30 08:50:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 08:58:39.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-30 09:10:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 09:20:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 09:29:32.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-03-30 09:34:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 11:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 12:40:02.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-30 14:05:08.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-30 15:46:02.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-31 09:08:21.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-03-31 12:24:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-04-24 17:14:30.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"}],"messages":[{"id":"b6ae8c7165b7dc18a18b07b2487c3d8027254e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-03-30 08:42:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afc3a2129212ec9e6ee48d10c69c2899b005e6fc","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-30 08:42:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/b29bde34153c49be8e21e4716aae34f1 : NODE_FAILURE Node request 200-0000689900 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"674b5ca0a22a6358180064cb1656223c67dc4b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-03-30 08:44:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dc3abacca94d5a0a90a4b75acad23c0b26d7a251","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-30 08:45:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/44e8f82f56054514a8a97c6bf8af0831 : NODE_FAILURE Node request 200-0000689908 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f516a24b900072bddb8559ca8f229a71e6420a57","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-30 08:50:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/57bf7449dcbd43d28a825ba3357b4575 : RETRY_LIMIT in 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f66cc3e61fe84cb591c9c640325186a5 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":2},{"id":"4a356f43776deeb890c7a543b241b78c9014e27d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-30 08:58:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/2d6ecbac2b89417ab1ed56dbff99c489 : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"48ed7ad40eb71f4fa78aed51b6ffb9610b8c7416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 09:10:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d550c467a92946b393ecd5306c563100\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a17cc03e8e55469ca7f3aeb0b0f94bb3 : FAILURE in 15m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/357b2eec4a24466cafad8b1782cbd216 : FAILURE in 17m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5824c6102d1e449dbd1495c86157850a : FAILURE in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6dde474fa513e9225c03b176dfa818ad3d3f6c31","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-30 09:20:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":1},{"id":"40d4cd141a5382f86a6b9b9db1c8f9d62bf2624b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-30 09:22:52.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":2},{"id":"3ac755917bd6d08a85f28dc0fe81cb5188c37338","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-30 09:29:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982620 : FAILURE in 40m 03s","accounts_in_message":[],"_revision_number":2},{"id":"33c59b030d436a6056b1892f7d19a0322b936dbe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-30 09:34:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3d61960cff6d4a1aadbe91589b66126a : FAILURE in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f865cec670e0a7bf42e2a3cbae439c639d6ebf55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 11:19:16.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01084a1b5f4f462dab00bf69637eb6fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ad35e6ab94424eb281f37a44fc3847 : FAILURE in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d2598b3a71415cb31adfd73721c7a3 : FAILURE in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4309db6840ab480d8063b37099035610 : FAILURE in 9m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9c105615ddb41b4846e0106a8069b6c : FAILURE in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04ee967caa7488d9266e08c74ebbe0a : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8e91bf423b8c4cf9913b2808ead856a2 : SUCCESS in 1h 05m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/376b08335a6e4f46a5a12328a05d98eb : SUCCESS in 58m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e232c886ddf24b0790371dc449b369a4 : SUCCESS in 1h 33m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcd17d4c4cd24f168960a4e6a48b3973 : SUCCESS in 1h 02m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce722872adb1488db86325d3ff5bc8dd : TIMED_OUT in 41m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13dced7056744e709b4c60ece1225998 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59396f3d0cd44560a8afce8cdb04d840 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47ead16cc7a9437b87eb3d8a39424e1c : SUCCESS in 4m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9269838da08d48e5bdad79201ae7b4a7 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9955935df4e642ce9999e8d626d0b400 : FAILURE in 39m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a34cf04df0fa4dcd89fd86de02958cb5 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3159f23c7ec54be4b897524047b6e563 : SUCCESS in 1h 42m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d21ced0fcf21420f8e7228191d807f8d : FAILURE in 2h 28m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1d7c16d124144c0ebde9da16789f429b : SUCCESS in 1h 47m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c6eb9f4e0d034bef98a4a002884eb088 : SUCCESS in 1h 55m 30s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0d4a093cbc2c4edb96db923a208b8cef : SUCCESS in 37m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/56a9ea39c08c4c528d2bc7b6679c170c : SUCCESS in 1h 21m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f3dd6b16479649f38a4967260466ec68 : SUCCESS in 1h 29m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d506e40af9114e329378c9a2babaf44d : SUCCESS in 43m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48672cecce6545bbb84eb5f3cbee495e : SUCCESS in 2h 00m 03s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/ac38417a0b1043fa9d41114146f6254a : FAILURE in 1h 56m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ec938e687cad4f4b8cae9d71db46fb8b : SUCCESS in 1h 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da2d5e9f7e042fe931e9768c1b10d9b : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fc8d1b986bdb7ada69c51be63335706eed703186","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-30 12:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3618/.","accounts_in_message":[],"_revision_number":1},{"id":"74c90e96f1e3af4f3f74a9d013ab3663af54a413","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-30 13:02:23.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3619/.","accounts_in_message":[],"_revision_number":2},{"id":"e62c3edd044e47006073bcaea5ad2c0abad3a9ea","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-30 14:05:08.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1269/index.html : SUCCESS in 94m","accounts_in_message":[],"_revision_number":1},{"id":"ecbae7386f96e6d00ffd07c8e4c85eb60c1b6d3a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-30 15:42:48.000000000","message":"Patch Set 2:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1270/index.html : SUCCESS in 97m","accounts_in_message":[],"_revision_number":2},{"id":"50848b99086464e5b4afe946d3e22e0552479dae","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-30 15:46:02.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-2/FC/ : SUCCESS in 7h 1m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-2/iSCSI/ : SUCCESS in 2h 27m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd19388876c81d1ba9491c6419f2aa24fe981727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-03-31 01:24:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"eb433501fd39b3a6715e9563e9cddcc56dc74d83","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 01:25:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a409dc2f90904b059c086e0056b840f4 : NODE_FAILURE Node request 200-0000691995 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"1a1b7efa289c38a6e9bd1c4e8d9221494b40af26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-03-31 01:25:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0f28553534948cd7d9b149e9d6810a67a302bda0","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 01:26:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ec985f6e33bd4a1d85d58a235aaadffb : NODE_FAILURE Node request 200-0000692001 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"54fb431318d6c41ef1396d1613ba53af9535dd8c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 01:30:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f6e7cb7ff2e64180b0cce8ecbca6e21e : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9e0e3dea1a314a6d82fc438f029bc339 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":4},{"id":"2aeefa9d883d349bcaa4bc1207c7d8fea7b3df19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 01:43:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51bac7e46b1945d9bc1d4257b4fe270e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd0b76d5dd094a2494fd74104f467e1d : FAILURE in 12m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/43baf43167ef4d9fb93fdaf3078e5ff6 : FAILURE in 11m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/091342e1967f49f9bbb8c1074bb90224 : FAILURE in 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3e396f277612722640cefce9577b1a2a89855447","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 01:51:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982620 : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"572332a0d8c90b315453358ae6226a8c0d70702e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 02:02:06.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":3},{"id":"7559a1ff7a2aaca39f46dd46c6adb4d3b32eb35d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 02:02:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/78580a72eff44781a03c6af3721f1116 : FAILURE in 35m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a9a7785c8bf19db2bffbb16a0e7846ad65e57be2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 02:04:25.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":4},{"id":"9ed2dc257fb21f97c22cb0ab2f20d1145f82999c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 02:19:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2ba1eaf30b544b2794697b629cb24f35 : FAILURE in 47m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9a8fe994080bc20e12b67f987e933f42eb4ce492","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-31 03:15:40.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-4/FC/ : SUCCESS in 1h 29m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-4/iSCSI/ : SUCCESS in 1h 49m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"5a2eee489418c948f0654c4e31bda4b82e3aaf4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 08:00:37.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b751ae2cd1a47a48fa4f9e4fd2dda2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf0238dc2294763927c4921d8892487 : FAILURE in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9947dd61fe2d4462855bd2cd0bda3dab : FAILURE in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b71522fe2f14b7d952fadc8765015a3 : FAILURE in 8m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/020218fe07c548609b8b67d40fc53cd4 : FAILURE in 14m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e920a029fe4685920a0a7cca8a0d7b : SUCCESS in 6m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4c67f436e2ce45d7bea0b73d13ed7b3f : SUCCESS in 51m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21dbaa3d3ead4174a47b6d800566c0a0 : SUCCESS in 34m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a2acfe1c63649308c5de47fbd720164 : SUCCESS in 1h 48m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0aa5a97412a4267825261623fcc5161 : SUCCESS in 1h 03m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e11b7482e19b4f0a9a81ba53865b1b0d : FAILURE in 14m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/52347abb41224db0ac0897a104e8d676 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d52bfcf5e645408285191b0430a607db : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a8d51c6d10640b898270427a40e965c : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8699437a48d84a7fb4960bde21b7bcf3 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37acd995eda84e758fbadacdb1b8aaf8 : FAILURE in 44m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39921c641ad748558c0bcd7386ee68fa : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4dfbdf0350804a588b21e2e4a95b0774 : SUCCESS in 1h 49m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/967bf612dc3f447eb9d39b9d13c774f1 : FAILURE in 2h 38m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e105e2795ca4874a3158ef64d865fd3 : SUCCESS in 1h 55m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2947933dcd6949ec8fc4d9adb51d26f1 : SUCCESS in 2h 10m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/75b4c5b1d3ed4eb6bfbe9a303e751755 : SUCCESS in 26m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47a36366c48240b5bd72fe2f984054cf : SUCCESS in 1h 13m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9333b514cdbc46ada2f73da59eb58dcd : SUCCESS in 1h 34m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c234b3e96c3a407da6800ce90b203b4c : SUCCESS in 46m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df7494a73efb4e95a0dba796bdf7bf26 : SUCCESS in 1h 37m 20s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/bdd5c672eb5c4560b54d5a37c3a81c9f : FAILURE in 2h 10m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/553d48bf97d04bd195955b9fa0ac311e : SUCCESS in 1h 52m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af77fb38415041fabd37fa44208a0aeb : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":4},{"id":"6830877bad11c3cd9cfaeebd3edd1791be8b1ea7","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-31 09:07:10.000000000","message":"Patch Set 3:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1282/index.html : SUCCESS in 99m","accounts_in_message":[],"_revision_number":3},{"id":"b2c4522768428372b34b8548b5228bcf5fda72b9","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-03-31 09:08:21.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a89ad778b291186290205727f1df7d144defd2bf","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-31 10:52:08.000000000","message":"Patch Set 4:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1283/index.html : SUCCESS in 104m","accounts_in_message":[],"_revision_number":4},{"id":"5837d1cb0ca1fe0592715e4ffdb0fdf6eef01a1e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-31 12:24:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"66912bb28ee00c3a5197421a2b528d5af36fdd89","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 13:16:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3631/.","accounts_in_message":[],"_revision_number":3},{"id":"1d541449baabfc82418e0d25c78ecf88688681e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 14:43:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3632/.","accounts_in_message":[],"_revision_number":4},{"id":"65e627716eda6b1229ff6a2ee8b016b7ce0f2080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-04-01 01:34:26.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8fa0b94e4bdd131897f6d1e58d9ccf5868cc8953","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-01 01:40:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a3dae6bf9adc47f28d35664092146909 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/663793fe1ec74a09977df2b936650dee : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":5},{"id":"f27c18493592e75caaf17eb0415a3ee0b4703fad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 01:51:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497efc740c114b09ba5023657a2bfc0a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc7909a27ca84f55892907887bed1d0d : FAILURE in 14m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/120ca10a902d4256be3372a3dae97aba : FAILURE in 10m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/281d11c8e2894bf5b9b525d024ecdd1b : FAILURE in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a4292d21089f624a7689246ca563122b30cef92","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-01 02:08:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/81ccee6e7d8a4dafa202888ef1503911 : FAILURE in 33m 16s","accounts_in_message":[],"_revision_number":5},{"id":"440e43faab3b307ab7d629f98e2338348674881d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-01 02:11:19.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":5},{"id":"440686a4135dd2ee3530eb104771675232bec5b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-01 02:13:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982620 : FAILURE in 37m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6c1f3165334ff2a0ae09701c709ccd7e64362789","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-01 02:20:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/df8f595cc4fe49eaaa9eea102084fb78 : FAILURE in 44m 35s","accounts_in_message":[],"_revision_number":5},{"id":"953d1c2b1a1eb02330a86c4e0d4b1511534500f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 04:43:59.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5743332f4075483c8c40e83659305c8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f137d9f562ee483c9c215524aa7a74d7 : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d13597a969245c2ba30cee5c6049fcc : FAILURE in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/912a88f5bde544fbb349f5f31a69ca60 : FAILURE in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/553f5a0f9537407fa23a3bdc943c2741 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4175ab3bcb4eb99c43b99011879514 : SUCCESS in 6m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/d44966871a044642b8b43aa8065284cc : SUCCESS in 1h 08m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5c14cc0ff124bd7b1885a9d1116bd6e : SUCCESS in 1h 05m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42ea02e1b7734c6bbe2f2e2a79a14c7d : SUCCESS in 1h 35m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5976e05f90d1481b9d5d50fe9a7adc48 : SUCCESS in 1h 01m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e5e290bd5e2d4add8f5c8886648b1f4e : FAILURE in 18m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2a4fe0490912423a9e98392b7836af6f : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4fea8ede789a4747ba15938d4fe583c4 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb16627cd5304c32bc8b0d53c586134c : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a89140095141421c986e96df43a00798 : SUCCESS in 9m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7294b94ad74449d0a2cd3bfb9a8f29e5 : FAILURE in 41m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1cd83b470fb6496887771428b03dda1b : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e431b882c8a9430da8ad15868ed3a735 : SUCCESS in 1h 39m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/205485124e224822a097e4308e498199 : FAILURE in 13m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40475b254f5a4a458779f7e646fc9ce2 : SUCCESS in 1h 36m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d1fc88ecb5004c119f45528eaaf684cd : SUCCESS in 1h 53m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/031be0f17cf344b6b46a83ba74ad2261 : SUCCESS in 36m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8adecc9032b7493b918e58ed6f6c0ff5 : SUCCESS in 1h 18m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4fb297232429480ca3d8c7d305d3a11f : SUCCESS in 1h 33m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b8da5651eb284e05b2c98eca2b54ad07 : SUCCESS in 34m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c13afadf98d749c485d9045a7a726fd1 : TIMED_OUT in 3h 08m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/4969f48f0c9e4653bc5600d4aa6c351e : TIMED_OUT in 3h 08m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/270ebb6d0ea84a258963a587c767c46e : SUCCESS in 1h 46m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/728755c7b7d247729569c24a51c772f7 : SUCCESS in 28m 59s","accounts_in_message":[],"_revision_number":5},{"id":"12a810564932ccb392111a1d1d47d8cbf442c4ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-04-01 05:32:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"858f0ee0086d6fe08ac1394d2d39d661ba4f8562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-04-01 05:34:30.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"e0ea2c327e89d3f3f6cb592d341d816fefe18e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"date":"2026-04-01 05:36:49.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"c06b61e986332591a5271b106aea7f559dd6245b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-01 05:41:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4a46d17844fa41feb2336f14da061d2b : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/9772dfb105214ba8bc1bfdee45bbf324 : RETRY_LIMIT in 27s","accounts_in_message":[],"_revision_number":8},{"id":"4ad5c2aa6bab925e31f114b9510cc109a7fe060b","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-01 05:50:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/12aeaa5ed89d46c78ddcce8487039b40 : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":8},{"id":"9f6c3af73391766108ae8beea80bebab6c4d9d86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 05:50:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0d4839f4d64b608b30915e270801b0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13c453823ef14440b7222ac55f91f240 : FAILURE in 11m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/729ce78c87334674b4a2958789e281e5 : FAILURE in 12m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0e93fbe90224456fbe5c1af37a8b8db4 : FAILURE in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2438d5bb5a0febbd36fcbcdd031c6c1a033ce5ff","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-01 06:08:54.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":6},{"id":"e8e3a84d5d8a32673a55c2cffcc9c2e305cfcc55","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-01 06:11:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-fc-driver build/05319f94f6194b0eae7922f0ee8154fb : FAILURE in 32m 50s","accounts_in_message":[],"_revision_number":8},{"id":"13946766c96d3755297b655b280cad0338fa4270","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-01 06:11:59.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":7},{"id":"7145e4ae36edbc7c9dcf3ca0baa600fd57913dda","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-01 06:15:51.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f .","accounts_in_message":[],"_revision_number":8},{"id":"56e1135d1d78ba4860ccf8b2b4131d24113d43bb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-01 06:15:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982620 : FAILURE in 37m 08s","accounts_in_message":[],"_revision_number":8},{"id":"2d09b87667fa5fc92425101ec273f030a15fd409","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-01 06:25:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/4d2f5e67e4774503ab1376f40c9b8e55 : FAILURE in 45m 18s","accounts_in_message":[],"_revision_number":8},{"id":"02145e7cd8ab3caabe5ce1d9679527c805938f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 08:00:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/339944fdbcdb4207a75c5ffa7a27e781\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec7a229d1254b188442a4cc91d216c4 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ffe422a215249e98e81f7c36c494d07 : FAILURE in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51aaa9fca60d44c2bc7d9d246cd78d97 : FAILURE in 10m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d1eb4eb5a6af47288e3694873479cb29 : FAILURE in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb80d01a319248f9ba582316b5e1926b : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/f8839ccc8174413e98c741fd79d2772e : SUCCESS in 1h 03m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8ac822fef5746de9802ada0213ab564 : SUCCESS in 1h 05m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab7987cc117745da866b446f99d27165 : SUCCESS in 1h 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6742a23c20bc4dd2a0fe89e7ec4454a6 : SUCCESS in 48m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4e1c4cac397345499d46f5d7bac261ad : FAILURE in 18m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d8bfecee909749e0aeb68b3810a93398 : SUCCESS in 4m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/647a35fc1013417e8186c97c092e99b8 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb5337f8c85a45f59860f4d5931dcb16 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5819e7e9ed9a4b3993e90b33a25353aa : SUCCESS in 5m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2491781ad92148a082db8892b14ac2a7 : FAILURE in 21m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5a1a0337ccd404cb53b7be8f375ed13 : SUCCESS in 3m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b1bb2a3a7b6c418e9b867e1096810226 : SUCCESS in 1h 03m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c71e02982f704f9c97133d7c07b76a6d : FAILURE in 1h 42m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b2bfd1d8503b40dfa4eebc0653de7f53 : SUCCESS in 2h 18m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2d33f009ddb54effad8924c0202027d3 : SUCCESS in 2h 04m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/675c76fdfa8f4bc0b3e9984f65250d50 : SUCCESS in 34m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4902e83bb90549b989acb0c8688dda24 : SUCCESS in 40m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/55ab49c7e0b14858a106758006b58233 : SUCCESS in 1h 45m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4fed81c6cd254b5f9f7779110b4548b0 : SUCCESS in 29m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/64458f2b35c048318d118b81dd1a6888 : SUCCESS in 1h 41m 34s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/2c6fb84d029240cd96466bfacef116ee : FAILURE in 1h 55m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c41004fac58f4d2fa6e9c984625c1959 : SUCCESS in 50m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b577dff0c4844ba3a6961348ea3a8c66 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":8},{"id":"faeeccac7b8a8147f3738783ba20021d84d677a4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-04-01 08:18:21.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-6/FC/ : SUCCESS in 2h 45m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-982620-6/iSCSI/ : SUCCESS in 1h 30m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"33d9259e3dde7865f3d5e7c1ce4b33d33ae9829c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 20:41:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3658/.","accounts_in_message":[],"_revision_number":5},{"id":"941d95082a3a2e3a8042696554f6c7b2d16ab309","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 21:51:24.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3659/.","accounts_in_message":[],"_revision_number":6},{"id":"6156c81f5b68655840176526402cfc8b41714a14","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-01 23:03:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3660/.","accounts_in_message":[],"_revision_number":7},{"id":"8e58685f06b89108f1a68b7fe6a73a3c6c5fa9e0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 00:06:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3661/.","accounts_in_message":[],"_revision_number":8},{"id":"5538c7346824f301d1bf32175c885d5ae76274e1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 17:38:09.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3445/.","accounts_in_message":[],"_revision_number":1},{"id":"3a694d392693b8a0ab20bf18ba459d19247e8934","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-02 18:39:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3446/.","accounts_in_message":[],"_revision_number":2},{"id":"c42bd01c37d698b8a4e563370a9097e8c4b23743","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 08:13:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3458/.","accounts_in_message":[],"_revision_number":3},{"id":"10909e9949e3aa0e1a5e55c899414554efc6bcc6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 09:17:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3459/.","accounts_in_message":[],"_revision_number":4},{"id":"da9bc4083b50d0a9ccc9f4dc6682b87d399570fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 13:23:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3485/.","accounts_in_message":[],"_revision_number":5},{"id":"6a7daf6f9a109518970bafa25c8e2ff5936e1d36","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 14:37:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3486/.","accounts_in_message":[],"_revision_number":6},{"id":"92a1e4f5be0449984fda2fec44848ac8862c390e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 16:01:17.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3487/.","accounts_in_message":[],"_revision_number":7},{"id":"838bbca1b7e007ab4a75c9172f2a7d3ac476ec80","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-04 17:00:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3488/.","accounts_in_message":[],"_revision_number":8},{"id":"4de18d3446060ae5a13c91879370017d8c517c2b","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-24 17:14:30.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982620 : SUCCESS","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a0d9c2de04c5fdcdf90d0f4a32de4161b82c778a","revisions":{"33bdb65bf8e084664fdfbd3aa9fef00e3d40da4e":{"kind":"REWORK","_number":1,"created":"2026-03-30 08:42:05.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/1"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:40:41.000000000","tz":540},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nThe HostManager\u0027s _update_backend_state_map method was only updating\nvolume backend states and cleaning up inactive volume services, but it\nwas not performing the same cleanup for backup services. This could\nlead to stale backup service states being kept in memory indefinitely,\nwhich is inconsistent with how volume services are handled.\n\nThis change extracts the backup service cleanup logic into a separate\nupdate_backup_service_states method and calls it after updating volume\nbackend states.\n\nCloses-Bug: #2035571\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33bdb65bf8e084664fdfbd3aa9fef00e3d40da4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33bdb65bf8e084664fdfbd3aa9fef00e3d40da4e"}]},"branch":"refs/heads/master"},"b4988a0d9931a1a8586d1fd4e9bfc45e40329b0f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-30 08:44:37.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/2"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"taemin kim","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:44:37.000000000","tz":0},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nThe HostManager\u0027s _update_backend_state_map method was only updating\nvolume backend states and cleaning up inactive volume services, but it\nwas not performing the same cleanup for backup services. This could\nlead to stale backup service states being kept in memory indefinitely,\nwhich is inconsistent with how volume services are handled.\n\nThis change extracts the backup service cleanup logic into a separate\nupdate_backup_service_states method and calls it after updating volume\nbackend states.\n\nCloses-Bug: #2144863\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4988a0d9931a1a8586d1fd4e9bfc45e40329b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4988a0d9931a1a8586d1fd4e9bfc45e40329b0f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ede11457343a4686bb04e0d2125c5eb0ba0b0b93":{"kind":"REWORK","_number":3,"created":"2026-03-31 01:24:22.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/3"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-31 01:23:21.000000000","tz":540},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nmodify FakeHostManager\n\nCloses-Bug: #2035571\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ede11457343a4686bb04e0d2125c5eb0ba0b0b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ede11457343a4686bb04e0d2125c5eb0ba0b0b93"}]},"branch":"refs/heads/master"},"a8ba7f6c4b374adc26eb8d92ecad5d234410474a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-31 01:25:49.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/4"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"taemin kim","email":"taemin23.kim@samsung.com","date":"2026-03-31 01:25:49.000000000","tz":0},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nThe HostManager\u0027s _update_backend_state_map method was only updating\nvolume backend states and cleaning up inactive volume services, but it\nwas not performing the same cleanup for backup services. This could\nlead to stale backup service states being kept in memory indefinitely,\nwhich is inconsistent with how volume services are handled.\n\nThis change extracts the backup service cleanup logic into a separate\nupdate_backup_service_states method and calls it after updating volume\nbackend states.\n\nCloses-Bug: #2035571\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8ba7f6c4b374adc26eb8d92ecad5d234410474a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a8ba7f6c4b374adc26eb8d92ecad5d234410474a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a3cbcc94b77ba26baab81931825e2d0bde7d26ad":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-01 01:34:26.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/5"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"taemin kim","email":"taemin23.kim@samsung.com","date":"2026-04-01 01:34:26.000000000","tz":0},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nThe HostManager\u0027s _update_backend_state_map method was only updating\nvolume backend states and cleaning up inactive volume services, but it\nwas not performing the same cleanup for backup services. This could\nlead to stale backup service states being kept in memory indefinitely,\nwhich is inconsistent with how volume services are handled.\n\nThis change extracts the backup service cleanup logic into a separate\nupdate_backup_service_states method and calls it after updating volume\nbackend states.\n\nCloses-Bug: #2144863\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3cbcc94b77ba26baab81931825e2d0bde7d26ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a3cbcc94b77ba26baab81931825e2d0bde7d26ad"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fcdd29d782a300ec0050275571ca2c58b41935e2":{"kind":"REWORK","_number":6,"created":"2026-04-01 05:32:51.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/6"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-04-01 05:32:33.000000000","tz":540},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nmodify FakeHostManager\n\nCloses-Bug: #2035571\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcdd29d782a300ec0050275571ca2c58b41935e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fcdd29d782a300ec0050275571ca2c58b41935e2"}]},"branch":"refs/heads/master"},"68034b8166f9ed4c477d4b99659bc1da53b2ea03":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-04-01 05:34:30.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/7"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"taemin kim","email":"taemin23.kim@samsung.com","date":"2026-04-01 05:34:30.000000000","tz":0},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nmodify FakeHostManager\n\nCloses-Bug: #2144863\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68034b8166f9ed4c477d4b99659bc1da53b2ea03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68034b8166f9ed4c477d4b99659bc1da53b2ea03"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a0d9c2de04c5fdcdf90d0f4a32de4161b82c778a":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-04-01 05:36:49.000000000","uploader":{"_account_id":36711,"name":"taemin kim","display_name":"taemin","email":"taemin23.kim@samsung.com","username":"kimtm23"},"ref":"refs/changes/20/982620/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/982620/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/982620/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/982620/8"}}},"commit":{"parents":[{"commit":"81c48732b13b4512ca63ac0fc91ccd7b0eec2863","subject":"Merge \"Cleanup glance image metadata during instance reschedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/81c48732b13b4512ca63ac0fc91ccd7b0eec2863"}]}],"author":{"name":"hyporia23","email":"taemin23.kim@samsung.com","date":"2026-03-30 08:37:11.000000000","tz":540},"committer":{"name":"taemin kim","email":"taemin23.kim@samsung.com","date":"2026-04-01 05:36:49.000000000","tz":0},"subject":"Fix backup service states cleanup","message":"Fix backup service states cleanup\n\nThe HostManager\u0027s _update_backend_state_map method was only updating\nvolume backend states and cleaning up inactive volume services, but it\nwas not performing the same cleanup for backup services. This could\nlead to stale backup service states being kept in memory indefinitely,\nwhich is inconsistent with how volume services are handled.\n\nThis change extracts the backup service cleanup logic into a separate\nupdate_backup_service_states method and calls it after updating volume\nbackend states.\n\nCloses-Bug: #2144863\nChange-Id: I6e8adc9c6c68b98bfa530de2b30d66c83f92f99f\nSigned-off-by: hyporia23 \u003ctaemin23.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0d9c2de04c5fdcdf90d0f4a32de4161b82c778a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0d9c2de04c5fdcdf90d0f4a32de4161b82c778a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
