)]}'
{"id":"openstack%2Fcinder~982731","triplet_id":"openstack%2Fcinder~master~I20267d3011e2c7de1cfa18e229fb545f4394e48b","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-08-28 18:37:49.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"},"37923":{"account":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"last_update":"2026-08-24 20:16:16.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"37940":{"account":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-08-28 17:35:19.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-28 19:26:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20267d3011e2c7de1cfa18e229fb545f4394e48b","subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","status":"MERGED","created":"2026-03-30 22:36:56.000000000","updated":"2026-08-28 20:44:22.000000000","submitted":"2026-08-28 19:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982731","meta_rev_id":"c7b6a9296dad3afb74f8a3365341513ac75e3210","_number":982731,"virtual_id_number":982731,"owner":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 19:26:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2026-08-28 17:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-28 17:35:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-30 22:37:46.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-03-30 22:43:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 22:55:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 23:09:33.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-03-30 23:16:08.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-30 23:16:11.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-03-30 23:18:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-31 00:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 05:29:09.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-03-31 05:39:23.000000000","updated_by":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"reviewer":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"state":"CC"},{"updated":"2026-03-31 06:59:58.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-03-31 10:50:58.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-31 16:37:29.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"CC"},{"updated":"2026-04-02 05:47:36.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-04-09 12:59:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-04-20 16:17:17.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-04-20 16:56:52.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-04-23 18:57:39.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-04-25 17:50:27.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-15 14:51:54.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-05-15 15:56:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-05-15 19:29:03.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-07-31 17:13:01.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-31 17:39:42.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-31 17:40:36.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-31 17:46:27.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-02 04:22:54.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-21 16:51:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-08-24 21:39:06.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-27 23:57:38.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"94f1efd3d95eaf7ae54411c59d8e46e0ba5a96ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-03-30 22:36:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a92a744172ecac0f5c5c8b573227bda4b52682ac","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-30 22:37:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/78f35fcbff4f47ae9fe9f1158893a415 : NODE_FAILURE Node request 200-0000691631 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"be68f8aa89348c1bc6eb5ed218c914651d05186a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-30 22:43:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8fab216f734c44f3851c9b3de8a6447c : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/948ad00e12a64b8fba7aaf672770e0ad : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":1},{"id":"51a54bad5234c2cb67ff27ad19f615794717742b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 22:55:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8155f178c67f4771bdabb712083d44a0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9b40f775aa64563a50e73384220a148 : FAILURE in 10m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2ed118cf7d0e4c94a83036ca75e729d7 : FAILURE in 12m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9aeef6aa6a334fe38fae248167879753 : FAILURE in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2264c3d3aecea44c9ee78b26bd0ecccbc700e027","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-30 23:09:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/19f93febd97c4a70a1633b03ae50962a : FAILURE in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f1651986dd5537d1b560de0cc09445ef1562751d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-30 23:16:08.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":1},{"id":"58ef10d8ee67f9d70d3941ebe49b79b62b3a3ce8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-30 23:16:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"52adc5376a28aab5fcdde446fa23d4bae22da7fc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-30 23:18:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1fdb6ab271734eaf8625e98b519bdf12 : FAILURE in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5b41f2c340d450649ad128f0ff3073df64ab2d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 00:58:45.000000000","message":"Patch Set 1: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb85956b0aec4c0384026cc0a82a7166\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c850104f4b034982b7764a1f14854a05 : FAILURE in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb9cdd95c59b4b0290f0d11ced42e49e : FAILURE in 12m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6215854365a845be9629d5d8d58dec9b : FAILURE in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4f27c92dd214069bdbc6a1e05a17fe3 : FAILURE in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c562267fe25347ee88f73c9a24ff2c72 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/63fdc75460c4484fb33dba1022a95d38 : SUCCESS in 32m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41d34317ddf44fbda929e21748ba77ca : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/984a1dc8878344f6a25534fc5a537415 : SUCCESS in 1h 41m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9daf9d090b5458b8cabb08395a30b2d : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c0b6091b5e4429a45d8d580e7b8d67 : SUCCESS in 23m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a102c60d14434612b00e65635320806e : TIMED_OUT in 40m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f7d9338ad4264966aef35e4e00034179 : SUCCESS in 2m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e84f86874ca4b229228bec796bc3557 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e62026747ef49c38e6a2364d8ff38c5 : SUCCESS in 6m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/970cf1eefffc4dfc8cd839dd548ecff8 : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29bf5243d9a84bedae89512c6a96d7e8 : FAILURE in 58m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/679dc00114654416b3082a3902e084ed : SUCCESS in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc00f8bc71704116b4b2d6785b05d163 : SUCCESS in 1h 47m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2d2363afe5df44deaddc4c869f34d49f : FAILURE in 1h 36m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e34873632f0549c48c33f9888c67df9f : SUCCESS in 1h 54m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d4b7f672eabf40d9adcaaa6c81741233 : SUCCESS in 1h 57m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db0580d5b443422fae1ea154e0ce0f0a : SUCCESS in 34m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/911ebb3f73c248e5a815c0f1ea0398ff : SUCCESS in 1h 14m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f50e9f560ea94155a669681b5dbca23a : SUCCESS in 1h 42m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83c28bc9146b4b8fb1b0615b6f632de3 : SUCCESS in 45m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9f3c866ee7974d2a8da9b54f594c0870 : FAILURE in 2h 15m 29s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/3e47398b0a3242c0a8d39d140ee68649 : SUCCESS in 1h 06m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5dc7063ad7234538b51d0e81f21f120e : SUCCESS in 1h 38m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dafdae78e01b4bd2b54b5c5c73ac6d12 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bf4d2b5b51264135b7b40b74105ca547c6855888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-03-31 03:28:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"418ab8c956d9ddc3bd871a253f53d046f41bdd33","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-03-31 03:29:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7f1e2347b49946238822ec50191fe80e : NODE_FAILURE Node request 200-0000692248 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"5a6d62fa2097ac8bbdefdb10a091fb996ab7f5e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 03:34:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/47cab7abcdd84796ac0070e83b4d597f : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/87ae2acb9d55431c806576929b5c3039 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":2},{"id":"db17912d289b7a4ae7c1d7a47905f18ecdeb74e9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-31 04:06:08.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":2},{"id":"9c7f390682fae94af0eeca82c3ea395fd7b5b826","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-03-31 04:09:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/9a19b35a904147349552056a240f040c : FAILURE in 40m 15s","accounts_in_message":[],"_revision_number":2},{"id":"cf800276c619f28b0ff35e7bfca43d90f3d72610","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-03-31 04:10:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 40m 22s","accounts_in_message":[],"_revision_number":2},{"id":"36d3f64f93b9cbfe1d154674af66eb5a537f0697","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-03-31 04:10:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c59083fcc31f4072bcfbdd22093d387f : FAILURE in 41m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2aad21d3a3dbf742a2a82279bf64c1fcb9d4a76e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 04:24:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f6748c6378946b98a457c06753deebb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ddacaf7c38e43f099d49977317baa8f : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7eb4c6537edf498ca3410700081573aa : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b858e4704fe54151b46eee55834c5c37 : SUCCESS in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89eb3574bff652d8747591def476739e9ed356ea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-03-31 05:29:09.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-982731-1/FC/ : FAILURE in 6h 51m 51s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-31-982731-1/iSCSI/ : SUCCESS in 1h 34m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"df1960ecf841376a416f8f74996bc9e77269a81a","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-31 05:39:23.000000000","message":"Patch Set 1:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1280/index.html : SUCCESS in 103m","accounts_in_message":[],"_revision_number":1},{"id":"df0cb358d704812ac29544f91eb1964036f6224b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 06:27:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/586b8eba02f64f45b84572846b22bfc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c52fd211aa4b3c96d7af9f9836cfba : SUCCESS in 8m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67b683d19013459eaf7b5ef11baa6493 : SUCCESS in 10m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fd05f923e0c4f888a40e985cea6c600 : SUCCESS in 6m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f9f2c311ee7149bfb8caa4f19ca8f517 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9055d7afbb043febfce68c47d33676d : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8e9ff78552cd42868aa87e0356d279ee : SUCCESS in 1h 02m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae89f9f259b843a2ab287c6cdfcc51d6 : SUCCESS in 32m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7d2645691664067a0d25ab22e4ef1b1 : SUCCESS in 1h 39m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f364e4fa1549109311a8b95d5fbd56 : SUCCESS in 36m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/457ff2a2197c4ed39ebd5e704db0aa83 : SUCCESS in 19m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/abf1f6c78b4c490e8800bcf18758bd73 : SUCCESS in 17m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e03cf1962faf42b8aa60729adbc1bea1 : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c57ac128c254ab3917f7fa2ba32dbda : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/516be34e16b64d8db524010aeba676d1 : SUCCESS in 9m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/deba87ad24f9458c810744b039245ce1 : SUCCESS in 8m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b79088ba673c439399665ec889064acd : FAILURE in 18m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96bab79e30c447ed8e4f87f3f4a184f4 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/62d81f09e647486a967a73d1e9359799 : SUCCESS in 1h 42m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/261686cb22514cb1a8c77fd5f2dd7e90 : FAILURE in 2h 52m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d5e2a6c063d4defb0050d9f3fc6d016 : FAILURE in 1h 57m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2e4330278f954e19b76b9ed5ef7a730b : SUCCESS in 1h 30m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ed6dcb5b396347a6807f0f41af2a0f73 : SUCCESS in 35m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/74f9481ffd4948eb9e39615d4843444f : SUCCESS in 1h 15m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e88708b50e14ff3a32192d484e8137f : SUCCESS in 1h 19m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/458fc0920af94d99afb77071ae05ac8e : SUCCESS in 28m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f22d8511bf174d5cbbeb542f0117fa4b : SUCCESS in 1h 10m 16s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/a2f013fffafc425482b45b534c0389a1 : FAILURE in 1h 17m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/589f0872d1c74ce4a446866723e800b8 : SUCCESS in 1h 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4e7e506615047f09acbc548f69a6043 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d7cd8161bb6aee3ffca994bce6ce1698508fa252","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-03-31 06:59:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-2-fc : SUCCESS in 58m 45s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-2-iscsi : SUCCESS in 1h 7s","accounts_in_message":[],"_revision_number":2},{"id":"37dc5bc2e0203ba69002ac24b7cdf0ce0b103d2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 10:50:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3629/.","accounts_in_message":[],"_revision_number":1},{"id":"c56f780212b46433aca53c831b40f0501ed21c0c","author":{"_account_id":38710,"name":"VMstore CI","display_name":"VMstore CI","email":"alexey.khodos@gmail.com","username":"vmstore-ci"},"date":"2026-03-31 12:25:42.000000000","message":"Patch Set 2:\n\nVMStore_Cinder_NFS https://d3krvozkty9bs0.cloudfront.net/build_1284/index.html : SUCCESS in 93m","accounts_in_message":[],"_revision_number":2},{"id":"263569abc05c5b8354c190f3c9f2a789be598cde","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-31 15:59:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3633/.","accounts_in_message":[],"_revision_number":2},{"id":"7e6edac8e064cc6295bbbd3ade923537339da809","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-03-31 17:04:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26ea00e2bc985ffcfd3f68580cbbb1b5f1d8edcc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-31 17:11:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1d2f2fcd443e43e290d5f67dba8fa21f : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/7fc9bbc1bb964634acbc308d63a87fd6 : RETRY_LIMIT in 32s","accounts_in_message":[],"_revision_number":2},{"id":"0cdd92b8934225cef7afdd6402dd61307e5caaf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:25:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f15454dd8240d3b7b21bf83920e077\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d254aff51c6543fa92a059d29497c1ae : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52d8f9afc0134ff08088b38577d07010 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7e75b757c1b043e0b90ce7b0221e1e76 : SUCCESS in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba1b7f6a0dfe023553a6db896480afd126623ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 21:55:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa39ecb19d2431d875a7eb56ebd4b15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15fa80811bd49b09d071252877b0062 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84e900977bb54b7ba1e3a18fda9b9b25 : SUCCESS in 11m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf5fb286f644ef9b4ce975609555185 : SUCCESS in 10m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3378fb278d9e4c68a17f43e9aa3095dc : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b33a8f3f3c481489417192e60ddfdc : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f20a70924b5e47758786d203687ef96a : SUCCESS in 33m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aab41ac1f8dc4b76a1598b29910849a8 : SUCCESS in 34m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fac25a687a6c40a293d5bb2eec74ecc3 : SUCCESS in 1h 48m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92209bf861534ddca4634d966105a4f2 : SUCCESS in 1h 04m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8709130505014e4f9494ad15095c4709 : SUCCESS in 17m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/edceda0584a84b5a929452e07cd75c71 : SUCCESS in 22m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4faa1d8b265d450eb201e06ac4c14d7a : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac7b3699a07d4561bb23a15163c8e886 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9d424ff721f434884e354a8934cca93 : SUCCESS in 5m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/47c22d1b240740fca8c4868ac745f70f : SUCCESS in 7m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3bbc2ca6f1ac4d33a3753fcd61d3cf85 : FAILURE in 46m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/987c3bf795de47f1bc1788fd3e861806 : SUCCESS in 5m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d81817954c50410aaad83199bf7b3fb0 : FAILURE in 23m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/acb856da7a3049d9b6d622690c35bdac : FAILURE in 1h 49m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d30d63959d14bb1b3e2f33f375243d8 : SUCCESS in 1h 58m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/94f084feacda489dbaa5939f1bed14c5 : SUCCESS in 1h 51m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5eb1775e409f420d9282029a34cebf27 : SUCCESS in 40m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a2bef6476239412ebcf4fb3b079910e9 : SUCCESS in 39m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b9f68b79013b4cf2bace380e1f4b372d : SUCCESS in 1h 32m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5c5e35181f2e4d80be6debb4b29a336a : SUCCESS in 42m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/27be1581138347d996ef17c0422464b2 : SUCCESS in 1h 07m 32s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/99f902c237364c8486c71279b68bf753 : SUCCESS in 1h 05m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b5754739d72404484b209e162919c86 : SUCCESS in 1h 46m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50e6d7a0faa74444ad7720dd06ab29b2 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"78afcde4e056843c606c6f8fba43697a780f9fcb","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-03-31 23:16:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c8f2c9a622066deeb8082af2497bbad4a5712d0","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-04-01 23:32:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d60ae6ea957a04542fbe091da8da5e80e4f39d3b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-01 23:38:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f58ae074b5eb4a1fabe15e8346bf47b7 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/b898f3f62bdc4f72aab4def148f1d0d2 : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":2},{"id":"8536d49c7ff2f29d39b856086f2c49fa54407816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:41:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e22324cf1034601a72258ee694d44da\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9116f6b723f64f86a153f80cb555f665 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/423d325f473b4299a7ae9657d6e2ad96 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d1cfd87e581445d9f3b51bdba5099e6 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90318b5a50c3440198331ee21e6ffabda2a29615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 05:47:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6a5ee2e03514573964b7e9bd6df1b59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c753942a3d6f4005b11c85a80caa15cc : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3281e27c5bb349ae9e8bc1483c40621b : SUCCESS in 12m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8568cc9877144c99b94a6a991b25ecd7 : SUCCESS in 10m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c153a9c965de426a9a4943edebbc11ca : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0808338a804953b77ee9c505a92be5 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/5e259e7e2ffb4519b0954f83f393b036 : SUCCESS in 59m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/920481bb4e9a4cdf8d45265f040821a7 : SUCCESS in 59m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/859fb8a1797041fbab5b8da1debacfc9 : SUCCESS in 2h 03m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31fd14a3818640e1b620c35d4329aeac : SUCCESS in 1h 04m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3094dda1db104551a0b21288752eb17e : SUCCESS in 22m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0cdac4481bbe43b995bea93d2c48123f : FAILURE in 26m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fb33cf0536a14de4aac4527fecf672ef : SUCCESS in 3m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f0ff0a0c13d41a195043e8a1a28db23 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ef910c4e04a44bc8fb4329c56bcd167 : SUCCESS in 4m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6986ff0a858a4f768d99bfb07bffb8fb : SUCCESS in 5m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bf3a492c115b48f18a2350205b2b8d6a : FAILURE in 21m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/492cd62dbfeb4662aa9c12ef88b5bca6 : SUCCESS in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5114ec46169a49ec9118574b8e6481c5 : SUCCESS in 1h 04m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d15a0702484645a1806ef3da77c75bb9 : FAILURE in 2h 43m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/932a4f1db2844c29950f731597cdf827 : SUCCESS in 1h 59m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cc67d24e86e94dbc85687823bbcca53d : SUCCESS in 1h 47m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3a92c6f4c844450494aff9c56b6e9ae7 : SUCCESS in 19m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/497be13cb8244364bcad557d4d5e1208 : SUCCESS in 40m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/69bc6a2abe5f41819336234126f2059c : SUCCESS in 1h 31m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bd16cf8d8b3f45e7a67b3f464cc7e02c : SUCCESS in 49m 52s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6327ff3e4cdc4f25a6924fe60266be64 : SUCCESS in 2h 07m 24s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/8be1b9065fe54af385e8a4ccf3b2e51f : SUCCESS in 2h 14m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/251030399dd14898a84a834765d6341f : SUCCESS in 1h 45m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95290a7e56c44c158ab69b75864f7e34 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1e21dd36a83ba5b1f800784b450d442da83a5d86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-02 05:56:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_04_2026_05_50_33\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"9b21890cdaffea7932c8e9e89a2de07058486236","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 05:54:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3456/.","accounts_in_message":[],"_revision_number":1},{"id":"3ddb77e3534a0da284232ca149eb1c28a323de8e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-03 10:20:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3460/.","accounts_in_message":[],"_revision_number":2},{"id":"87430d54ac293ddf8810164fa834a3f1e8264063","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-04-09 12:59:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c3653619db50f80a3a29c4415e56c601d216a8e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-17 23:19:06.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"caf3efe03b59273048d98c720413cb1c221048b6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-17 23:24:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/70ed00a9d5564496a59d8be74f7ab982 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a4c47993f7814663a4e10751b214f626 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":3},{"id":"9063690f59f7389f465f7197fa700c81a5aeaa7b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-17 23:56:19.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":3},{"id":"6294946e1de8332906dbecc42487464774b3672d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 23:56:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c26a420e28e74ead92fde84d05588a78\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1dc60a9240548e592f1cc3306aa1893 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53f430de4a504d8cb35151630e6b2902 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/81d0a932a44a4d8e8f104ad61aa15887 : SUCCESS in 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6cba63cc1ba2e100485f0f0dae95f1e849270a5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-18 00:00:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 39m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b954fd16b3640d9a20bcd92a1aa897c17ac6dee7","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-18 00:01:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver build/b9d6b2770b074c7d8ca76d427d87eeef : SUCCESS in 42m 06s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":3},{"id":"492450182455b4f0d8c145413b0f3234f4531a13","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-18 00:03:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/100019a01c674f6d84a4c6bee9a321f4 : SUCCESS in 43m 25s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":3},{"id":"b1f44430f6752ab1820467e7e955a7fd7c317d52","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-18 00:06:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/4ab70565e562407a9b62af1cfacb2e12 : SUCCESS in 46m 21s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":3},{"id":"4315438b43d8cce069132b9cdb1cc54d24affcfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 06:04:15.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb9da4de026e4d04b0d2f2dfbb59adfb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f060b13dd5c748e3a97706ec0d0f100c : FAILURE in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ff3d95dc22446c39c959c1fb9f075b9 : SUCCESS in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/692e641907264ca48c6b1f43b4905804 : SUCCESS in 11m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e72cf38c24244f3a0e259ae64fbe2d6 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95271a36386c44dd85e40a0ffa5d1aa7 : SUCCESS in 6m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b3a187abf8d0463e8480125274c49d3b : SUCCESS in 29m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21fac67c731d4279bff7422411e2de24 : SUCCESS in 34m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9586a8920a404d31a570c1f6beb32bcc : SUCCESS in 1h 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/774b1122d96d41daa14c5c3715dd3c5b : SUCCESS in 1h 07m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bd1663dc660446cac3aeef589626a57 : SUCCESS in 11m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb4993f56bdc440d9e058f7ea534cc4d : FAILURE in 25m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9cfa66cf92d54763903e0ff2781760c3 : SUCCESS in 2m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc5042cf6ca7448cb3535c6f0b899091 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9119723b08b548068bdf22477f8e9aab : SUCCESS in 6m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/12032fa7f2fb48ba99b1834798a7e9aa : SUCCESS in 6m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7e79dee8f84b4d61afa44b8f4aa82c3f : SUCCESS in 38m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a44ddc755ec54f51bf0e802f61aa8f76 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a669be084bb44443af929c275c9b7f8a : SUCCESS in 1h 47m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8ce3bed4df054f3b8e80c92d32bbcd97 : FAILURE in 3h 00m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1dc5d5ca632f4072ab9fffb030ea3ca8 : SUCCESS in 1h 54m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f746b9378da34be6bd12df49918aa9d0 : SUCCESS in 1h 25m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/464a9747065b4af1927bde5d41a7c538 : SUCCESS in 34m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa1a515967a2476cac0bb73db0eb699a : SUCCESS in 53m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/71a882d6531041fd8d278a9c42d90ee5 : SUCCESS in 1h 45m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ff309d5066a49a9be1e911755b14ac3 : SUCCESS in 31m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b9fe2422f014c1bb55e3ba4e26ae8d0 : SUCCESS in 1h 25m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/44141d60d1c7496485b95fb937633a15 : SUCCESS in 1h 43m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0edea1b69f564df7931856b49aa9e62e : SUCCESS in 47m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8ecb21fadd404cbf94afd45a1b182a2464b410a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-04-20 16:17:17.000000000","message":"Patch Set 3:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/1547/ : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"7271524498986a46aa077cfa941bd2880d2169c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-20 16:52:28.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"e688de7b0118d10c5956b173b05b05127ca62921","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-04-20 16:56:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9b3853a14fb5c2bd5ce6edaec43ec94b04b5d82","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 16:57:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/554600b391af4300bd1108e1b3317d5f : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/1e3645235c1a434583f5ef681f4ffe86 : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":4},{"id":"8fe9340ee7ad2ca1b1f74f1c081da4ff4b74ca13","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-20 17:17:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/92d8be47db364d5ca5c219706a13d17d : FAILURE in 25m 00s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":4},{"id":"896683d8d7f134bfecd51c4959be397a56df765f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-20 17:18:10.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"3ab391076ab79df79de30a64643897a5f1aad262","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-20 17:18:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e9d63663c293f7eb71d1714805d89b2659e97dec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 17:23:09.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b87c4cd6e99442c4958f892c138da6f3 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/e82ed3a671dc4c91bc8db1021bcdb5a9 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":5},{"id":"e427cdb936bade5026950761b05a68ea774331a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-20 17:31:13.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":4},{"id":"b4f1aae8abdf2be21cbd65ed8808ce27f6fd296e","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-20 17:42:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/a76f632a6c344408bb734f87bf6a4f65 : FAILURE in 22m 28s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"1b930d4a47c603a2d8a0c03920d35c35b95c20de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-20 17:56:15.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":5},{"id":"77dbe82a92f93da7b31336725688570245312078","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 17:56:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3901/.","accounts_in_message":[],"_revision_number":4},{"id":"d5be84fcd70d0f11c4a115830cbd46dbd399a7b2","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-20 17:56:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/2d777381edbd45db9c3436ef8cacbe27 : SUCCESS in 36m 59s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"8e39d26fc5f2cc519a6c0d47986888172d1eea99","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-20 18:01:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 39m 02s","accounts_in_message":[],"_revision_number":5},{"id":"c1068777b6e0dd1c9d0d6117170e8c570fdc41d0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 18:02:36.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3725/.","accounts_in_message":[],"_revision_number":4},{"id":"e74ba2be872233c69862dfb3060c928dc3c5137c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-20 18:08:26.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a5fa4c3d9e694d8f8d267ab817bb8de0 : SUCCESS in 44m 49s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":5},{"id":"360e1014db3ca448b43cec3c2818e05cf156d8b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 19:03:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3726/.","accounts_in_message":[],"_revision_number":5},{"id":"8719ea18056c3fc365e67779547adac387fca68c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-20 19:08:39.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3902/.","accounts_in_message":[],"_revision_number":5},{"id":"7701094b9a4361607aa70ef3d2b1b9963e0de350","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 19:39:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/326805f1a53a469e9e57658f2ee72de1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b1c4106a2e240f0a48801e834541acb : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb7fa1a1f39644d488ec99513ed4b1a6 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c7be381d87714bb58582be1d25c3c527 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b31b7ff61e4da0d3c7c2f5a639fec09cc48b8d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 20:58:15.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1f2e4b798324a42955fde4b2f5e19ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbd1e53c9be4e339a962294ea48e31b : FAILURE in 8m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a36097afc444294a8f3586dfad90730 : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e759cee4ad51400483b9a99412035261 : SUCCESS in 8m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6088b87969b8407ab9f4c7d8ada28e2b : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82411368c9d4b92911c137f959621af : SUCCESS in 15m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ebb3bda891c1450dbe7a3284cab25638 : SUCCESS in 55m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b42496527236456189c52476f25c5de2 : SUCCESS in 1h 08m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/359d4b1f8a554d7c93ade0a7bbe11e99 : SUCCESS in 1h 30m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/651f365cc82c4a0db00071d54ecb7b5a : SUCCESS in 40m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/847faa648efd4841b00ef47022cbe3dd : SUCCESS in 15m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a049702103aa4bb7b2509d97cff5e5d2 : SUCCESS in 22m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/beb0f81ddc4543aba36f67db8cb62f8a : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88f391a5e80447538ea5cee014e63ae0 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7888bad28b86436ca1ae81a6ec94835f : SUCCESS in 8m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/79ba590bad75483482df4c545bda46b1 : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0df2e6a86bc24ec9a9729654d9d50f50 : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2cf9603b0a69431c8d93798897ea4355 : SUCCESS in 7m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74a39f88fb974ffba3773483b21da27b : SUCCESS in 1h 48m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/720f752823b84f0c9e8676d5c0d75fae : TIMED_OUT in 3h 07m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5407a7e2dcde414889024ea591700f5e : SUCCESS in 2h 11m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/85c6b5cdbd3d4b6a8755556faf392cad : SUCCESS in 2h 00m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24f21cc3ab5f48d08b0d0144ef00383a : SUCCESS in 27m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2319400236964d8d866a6926dd6afa0a : SUCCESS in 1h 02m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6a15b94394c8433f8025960704cfab15 : SUCCESS in 1h 33m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8965cba546f742c691ba6bacf8aea348 : SUCCESS in 25m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eac5ce61d0d24c48ab1e65fc217d93c1 : TIMED_OUT in 3h 34m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fdd10156ffc7436c884859c84284f346 : SUCCESS in 51m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58d999503b0848bd87d372592521c434 : SUCCESS in 44m 54s","accounts_in_message":[],"_revision_number":5},{"id":"dadde8de1e90ca118743d3376d9ae3395b1bad8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-20 23:12:03.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7b6b5bb2b9624133919d1089d7f3b09f07d1e037","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-20 23:17:04.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/899a2293a6d948ac9183d58a46061fe4 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/803dee824c6747e198301bcd423e0bd1 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":6},{"id":"0d1e3949efdc1caf57cb369d07ee2bc9a54692e5","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-20 23:35:32.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/efe131cf1e7149e29eda6c2f5d98d0ba : FAILURE in 22m 58s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"4d45df8888105a56603ea8b4a1f18061e0b175de","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-20 23:49:37.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":6},{"id":"881f7e3bbba7c688748ee59a90c9a2126481496d","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-20 23:50:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-fc-driver build/fa51216734b644f38f39479ca27fb70d : SUCCESS in 38m 15s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"95d9f9ec3093b3a2ddd63fba5475710dd0b088a2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-20 23:52:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 39m 02s","accounts_in_message":[],"_revision_number":6},{"id":"47ee8ae0f879b6cb77a1f364aaddcebe73e30cc4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-21 00:01:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/0b9c1cce0bd34ad387ace3771874f1b9 : SUCCESS in 48m 57s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":6},{"id":"8fbb7eb836720ba296dfae503bc2c74945e28f2b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-21 00:22:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3730/.","accounts_in_message":[],"_revision_number":6},{"id":"92989931d47102eb489bf32f412d844915f69303","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-21 00:31:07.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3906/.","accounts_in_message":[],"_revision_number":6},{"id":"2819b833716faae61f37f4f4388783ffaa2b3e05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 01:09:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f3c53dfe06451bb1ac30ac89875b24\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31cb5e30846c454abb7933c8e361e29e : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a430da45ce474478bca119528adcf2f6 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f77b034f9b304d8d96d29b91055eba13 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e9f51198bc45002c67b6436d9b5d3e821db24bca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:34:17.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5741bbbd14f24c5294e3e0d8b9a2bcd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd14b11e914426c8b436fc97fe17dd9 : FAILURE in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f203f0b57264b6694d0194f33325eb4 : SUCCESS in 10m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abb98bfaab034b86be572a0cd66677d4 : SUCCESS in 10m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/655effce5f2a452baaf1d6ba80a43d31 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790506c2b6084992b33c5a8ff2b390c8 : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/058d282bd02b4ccb929f5e4b97005ec5 : SUCCESS in 1h 13m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7238ea0bf78f40f79b13a560bcd2ec2d : SUCCESS in 1h 05m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c414a0fd9e0f496bbe77cd76c61128c7 : SUCCESS in 1h 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67881276445f441981d3c807d99085c8 : SUCCESS in 34m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d775a4c1e22442d9916a6ae694d3697 : SUCCESS in 20m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f2283157204499d897d3831fc7d4320 : SUCCESS in 22m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e04b332097244dfb8faa2c8cf02659c : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/97e104959a0d4925964fa13310662973 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f0ec6c45ff54bf0a3de0ee49f5eb1cb : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8c7335185544aceba60ca3b0e1fcf6c : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/582e8aba6c814dae93c774a43cf3c3cb : SUCCESS in 28m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/501763c81bdd4e0584ef04a6f29bddf0 : SUCCESS in 3m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ccc69e309014203a4fad7eec43dc723 : FAILURE in 20m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec9ed07841ed403d963ccc2959b76dff : FAILURE in 16m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/242c52d217a140319c52585b6d4600f6 : SUCCESS in 1h 39m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8dc94f85b90b4aa3b70b7d4222a7cddb : SUCCESS in 1h 53m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dff019fc951d4520bffa1cc0096495e4 : SUCCESS in 42m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48a6f4c77fb74dd2a2a81e1ee04e8b67 : SUCCESS in 1h 17m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5e16f613dafa4d5892088d3db8ece359 : SUCCESS in 1h 17m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2ac40c67d4a641f792a5a417a80a6020 : SUCCESS in 52m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5097924c8bf4c65b5a9b10dfffde2cf : SUCCESS in 2h 03m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/804b6cb35ed449f491ceddb24bdeca9c : SUCCESS in 1h 50m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e2329d3e6ad4e899937719c7e873bb6 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":6},{"id":"e51322feb357e86b11b848cbf6d5142418e14dfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-04-21 04:06:28.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"657ee379589dc46154d1f6d21016f37448e5fa4a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-21 04:12:02.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/49ad3833266c448dbd1a44200de43414 : RETRY_LIMIT in 28s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c830d5db8b974662a1c1208b52e8417b : RETRY_LIMIT in 31s","accounts_in_message":[],"_revision_number":7},{"id":"ee09849c71e80ba9b9206fb6f68c7768403bd77d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-21 04:45:06.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":7},{"id":"203f73ac20f52f89f82716c121071c6824a614d4","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-21 04:47:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/ea531b8cb7e140e9aa79eb3015066ff4 : FAILURE in 22m 38s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"d91afd6d7e667b4ce18aa87d59262843d0000d32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 05:19:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7666d2bb0f4f0ca3d1ddd1669deeae\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f192a0179674121a4ae06f09b3ff8b7 : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37b0a019b9cf4decac91710180f826ba : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c3aeedcdbc643c4a0d8334b75340ca0 : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dc216f055d015a4845ae3756f8e8cfa6eed971b8","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-21 05:21:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hpe-fc-driver build/984be84994fb4019855ba31df43afce9 : SUCCESS in 42m 03s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"57bafd2c5fca9a8a82c7eda4de836b328323a105","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-21 05:26:06.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 41m 09s","accounts_in_message":[],"_revision_number":7},{"id":"ecad6b7aaf5ad9516758a642ef3b72a497ff9bb8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-21 05:48:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1c42c386c98a4f2db1e3f55c2b96de34 : SUCCESS in 54m 57s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":7},{"id":"7aa04a2e12b746f1e3c16943ffaab2db765945cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 06:41:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6cd05865edf460ab7ff8c4648daf0e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8368d0846074669905665076947ef2c : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90a98da99a2a4cbbb91061d8a9960ead : SUCCESS in 11m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c639607bbec7435c8b52c1040b2f5b00 : SUCCESS in 8m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/533ad7e16b764c2d8c48f1ab9a08beeb : SUCCESS in 14m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf646f4ac7ef4defa65ee469a88e1f24 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/a62a72d2ed044b7c8007242022d3c775 : SUCCESS in 1h 04m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73f1b67328c44c89a72d97f5ef5834b5 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86125f3a91e149978d1664d10478e269 : SUCCESS in 1h 45m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b064dafbf0243a98164076160d65888 : SUCCESS in 1h 07m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/333bd4450ed54272944912790ef44cf5 : SUCCESS in 13m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/131d047cb103495fa80dbd859f5d73c4 : SUCCESS in 25m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e893b72e8cbf4b769cb317f6d232cb8d : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2713c360e7234b52adeeffc3c7db6ff9 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1991d2f55944517aec5089fada999ac : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4f08d022e41f46cbac7acc2e00d42e77 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4aa0b2e8f3094b439c351073f9c2bd43 : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb9bf28fd25840738a1880d7bc58d7f3 : SUCCESS in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/07e13cebf7c84154aeedf35a10ac6105 : SUCCESS in 1h 20m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1854043458da4976bef2245090ec93a6 : FAILURE in 2h 27m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8460a93169e94726a1d8e7ef286035d3 : FAILURE in 1h 48m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0e50abbe336e4828b306fe58afb5a4db : SUCCESS in 1h 53m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d244cced1a00400097e68a6bffcd52f2 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fd5d3a5da0554ca18d05d32eea8b359a : SUCCESS in 42m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8d79999c873a47e98fbdb3cb08731818 : SUCCESS in 1h 32m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/260956610b1d4e149c9fe1ddfc833bab : SUCCESS in 40m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/907c37206e34448d8647cfc17ce63ac9 : FAILURE in 2h 15m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed6e0839c36d418684b712e872461878 : SUCCESS in 1h 58m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a986e249f5449af87a18728962d9fbf : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":7},{"id":"6c69cfca7f728984e94a85e0352c414f7b2306af","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-21 07:06:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3910/.","accounts_in_message":[],"_revision_number":7},{"id":"c710c63072a4bb340fb8115c416a1797c227e554","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-21 07:13:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3734/.","accounts_in_message":[],"_revision_number":7},{"id":"25bd3691ae0c9fedbd50950a75ed1581c4d98ecd","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-04-21 15:47:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"05cf1c01b1e3833567ed5eef9c978ea94519a98b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-21 15:53:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c38905eefbf9495db65e06269d87bfed : RETRY_LIMIT in 31s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4ab6c2ae6432424c8a23d78d03d5c590 : RETRY_LIMIT in 39s","accounts_in_message":[],"_revision_number":7},{"id":"e9951a5badba6ebfc2fd19ca06f6c61fe8f3e399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 16:51:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1102e0325ef345c8adc1407f2900676a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b2287b138f04c32bbe6e5e01c585e79 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f55bd7a386fc49fabec2356000642ad0 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c555368cd2947659d55330101d2e857 : SUCCESS in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"eb2c462b36082912f005f68d9995103351b118a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 19:04:19.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f93db356eff46119ee4244f101f373a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/017d97d0b8cb45049fbe945264ff1fe4 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb42b5c8ed2b44c0a0289222f4573123 : SUCCESS in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef50260a79964e6ebb02592c4bf215c1 : SUCCESS in 10m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1433d2adbea648f698eed09fc6a6f850 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840d5ddd0dba4580bdc0f2faf7fabc63 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/a28bae7a078440e698f3ba44cffe4177 : SUCCESS in 39m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89d10296af854643b03c48960ba214ca : SUCCESS in 51m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22a6961d163f458a857b61162ba64208 : SUCCESS in 1h 48m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/186d1e4612d54c7f8968cfe776aec243 : SUCCESS in 40m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4500bc50af9847a3b376f162dda57140 : SUCCESS in 21m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/26cdb0f55f7c4ffaba9cc8a52a19496f : SUCCESS in 24m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2209ea35c8964ad997d5ed0b65df6252 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a9acd9a903d4439a0c48f04290ea84e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b65a2e22d65448b8a17df7aa4fe68a7 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/268bd74e32114ecea48b3d44ec00cd04 : SUCCESS in 8m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26c22eda97e8463d84baa74c6dfc9d4c : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a91212c53644d81ad7fa80fdbd6e93b : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/775ad3c0ad9642c9bf316fafc21e8548 : FAILURE in 22m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/980cac47580d4333ba8b46afa80bb9f4 : FAILURE in 2h 31m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8364f071cf734559ada499b2bbe325d8 : SUCCESS in 1h 49m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6f59c6b2a79a49fd801c53b1fa128272 : SUCCESS in 2h 02m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/04c9692df82848e5b2b5f8f2f7d7a823 : SUCCESS in 33m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/60b970ad2ebf40ac90b37eede412298e : SUCCESS in 1h 02m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ced6317d118b4ee4b38f1e7d076f3c72 : SUCCESS in 1h 27m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/739f31e6bcfd45df904444bfaea48316 : SUCCESS in 43m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a78a6d8ca3094af98f0909adb523c928 : TIMED_OUT in 3h 09m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e5e8cc4a5faf4edfa5433c7231413e7b : SUCCESS in 2h 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b57cdbcddd8a434aa7d2a2d66d215f99 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":7},{"id":"30256e6b909fbbb54730e95fc85143bb70289d9f","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-21 20:05:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5f4f7deb35721522b52bece2734946f2525c05e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-21 20:10:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f4a8327dd0fc4f259156616a04402759 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/646a2fee82e34f8490987f48d4d25a15 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":7},{"id":"3ee8b8602f21fca1c5baae689cdbbaa3ba55f76d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 20:54:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c166135cb1f49a2a505036cb702c72f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96afedd39ca94744bb808d3b10f9e86c : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b18f6eb8edf4a5ba151e137969107be : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/73571ab5081e48f4a9152234c6abccfe : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b1c304e12d12df9befc110bab4a89bf829afc8f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 06:44:04.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42a650900b5a4341b00ce20b4862dd17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c373969e3f4a4dd48209b2258e27fab9 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/986ed5efbf864f438622413fa8f70467 : SUCCESS in 8m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e2796fd27874e948f5e1d001cdfb069 : SUCCESS in 12m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/250c0ec7bcf44682bcccc1ecf2548e3c : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f88f052f10045669c3c24f1545f31e3 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/4ce0db9b594d45dab87b8ffcce186058 : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a369036806c4110ba6dff33702c8f96 : SUCCESS in 30m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fefc52820ef64fb091cf1efe73eb78fe : SUCCESS in 1h 08m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb51b775ac67435291578408a353184e : SUCCESS in 36m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f90d1f7ddf004a6e9a8adb6b706ddc56 : SUCCESS in 20m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/febf18475d92402cb2b44227e85cc551 : SUCCESS in 19m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/34033ba7d9bf41688a06011611838dd2 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8416d3e121946109f956ada6e98d665 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3a993778ca3a4293a90a7edaa2ea325e : SUCCESS in 8m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/94744c96891b459e98e1ca705a1e182f : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8f7e14b9dee5414eb349fa60aedb58f0 : SUCCESS in 29m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ff714ef573b4c44b6d51e309db7f174 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f23eae900bd46348dc36fce91a2660a : SUCCESS in 1h 54m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48053be89daa45ada42b6ded05cce04c : FAILURE in 2h 58m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aeafbd75596e48a0b600523dd28a802a : SUCCESS in 1h 51m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b853c7f18edb4a78b1cca66e3d6f44b5 : SUCCESS in 2h 13m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6ee951b11882422fa7e18b3b41162d22 : SUCCESS in 27m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2dfd341e0ae3430a90e5498762ab61e2 : SUCCESS in 36m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/021deb6f52424fb380eecf3b7dc9ff1f : SUCCESS in 2h 02m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/743d1e952be14f8a8d3653927ef7516d : SUCCESS in 49m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6828420dcc643559e32aed3116928a6 : SUCCESS in 2h 03m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15bfb9fab0674ee19c0e06674716d153 : SUCCESS in 1h 52m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b95572e0bddd49599dbb76fdb9348d49 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":7},{"id":"b926abfec6314df97c799ba57ae78aaec571e8ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-22 06:53:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_04_2026_06_47_08\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"26f13ef36cc169d8f8b0ce77951e57b6caeed264","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-04-23 16:27:26.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-7-fc : SUCCESS in 1h 2m 37s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-7-iscsi : SUCCESS in 1h 12m 8s","accounts_in_message":[],"_revision_number":7},{"id":"16869028385c5736f35ec090da967a8f5e21f67f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-23 17:26:26.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d524e0aa104a84efe3e2c6a1ee8a2739957171c1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-23 17:39:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/834c792ee2e14f47abfeb96f27bd05d7 : RETRY_LIMIT in 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d69e6a90b13842c9aecefda527facaf0 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":8},{"id":"6ca319f2ffeb8eb0ae1d7db25ae96dfcd4ad4b51","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-23 18:07:56.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":8},{"id":"28f925b272ba7057a02996301600eef4617d2ca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 18:08:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0d3050746354a6ea68d1af543d89ad1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/212d964af6514f50beed3956551f19df : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/15b55adfb1aa4809a1b79c84c911cc05 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/076666d428a64c33925f71bf12f7ccd6 : SUCCESS in 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4e7084eb7026367f8b5879e0b4c174ea64e12968","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-23 18:43:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-fc-driver build/7cd09dc0f780429fbf340a818e770863 : SUCCESS in 38m 12s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"9a66659d212954ee8d3bf51ff5275334d4aacafb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-23 18:50:40.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 39m 52s","accounts_in_message":[],"_revision_number":8},{"id":"3668a5713ebb09074ed77ebf62efef69e51fb53e","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-23 18:57:39.000000000","message":"Patch Set 8:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982731 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"9e5d2ce5c5f0b6c709c01fc369a1cbf5dd75c7e4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-23 18:59:11.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/2a369a7c43564396b67eabf2a8d64c2f : FAILURE in 42m 16s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"5c36ae28290f0aeece95ae7c2ff1fd87a8e704fb","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-23 19:01:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/2b5ece779ac644baab9c3f54533ad8c6 : SUCCESS in 45m 47s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":8},{"id":"a839852218f6840088a160a52e8f855c632ab099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 19:49:42.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee996937b709470984f7b027045738ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29a52e75e06445f966d4782dfa18efd : FAILURE in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbe61e6c3f72483ea83e25d118eb24cc : SUCCESS in 11m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b4f5df196c24ddc847a168860bd32b9 : SUCCESS in 12m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae094a9f52a64def83079f5856f4691a : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d75e1e09c24b29bde47fb6ee517233 : SUCCESS in 14m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/946cafe173d249faad3dcf675b1c8372 : SUCCESS in 1h 01m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/544d49ee56894f5d951ed184d06c4fe8 : SUCCESS in 1h 10m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ace4145dc4024bdaa219a557676a96fa : SUCCESS in 1h 41m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c57618a48ff34bc7a319d75079a2690b : SUCCESS in 1h 06m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dc1222af8824b439bdf9705e109f577 : SUCCESS in 13m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7212e18e132e4aaaace53e884913c273 : SUCCESS in 24m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5473912a40d749ae82a5f653c6f075a5 : SUCCESS in 3m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ebe1f122ef604921b20f9f08c8888e5d : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64c283f8f3744243bbe0ec88a3b3f473 : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3b59aba8cc82465aaf6e12c9fa1b0588 : SUCCESS in 6m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ea93ed11673545fab2af188ad37d0871 : SUCCESS in 44m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7fdba3da5d44bf590c3dd27d7926a4f : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/244168f77c8f4dd99c969a73e78407a0 : SUCCESS in 1h 58m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/189d19d46c004ca19cd5179a387214d2 : SUCCESS in 1h 23m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/004f03f048ac4bcb820d46576b1fdab8 : SUCCESS in 2h 17m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/be0fd3e2d2074c53b0d6f03b82f98a03 : SUCCESS in 1h 57m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9f30f53cf7d84afcb619120a5d970b4d : SUCCESS in 35m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7013b3cfc479488d9edcbb4d344fa796 : SUCCESS in 1h 18m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8a6b24fb5274e0db2caddf776d31fea : SUCCESS in 1h 24m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7239f0b929624e91a86ed3bde410a89e : SUCCESS in 35m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b20a5a95109142a9ab7004e2336ec4eb : SUCCESS in 1h 03m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/26da2c4f00674b04bbde98edf6abf935 : SUCCESS in 1h 34m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/263aff7a1fad45bb90dcd8379a8cd554 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":8},{"id":"c497c122639b3b9936fa4d96d6ccee0b141cee0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-04-23 22:38:11.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"65560b87efe027cdac212bc14a79a95845522776","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-23 22:39:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3758/.","accounts_in_message":[],"_revision_number":8},{"id":"a92bd2e3ae7394ef0ca226b859dbca139979c819","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-23 22:50:13.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e0907723a99949349cc508c873329318 : RETRY_LIMIT in 30s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/8154f94145a54996ab12bebaa1fa1f1d : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":9},{"id":"63b350b00bfc6af74dec81f7da9007c55310bbd0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-23 22:57:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3934/.","accounts_in_message":[],"_revision_number":8},{"id":"7bef6087aca297f270062eb107aa4c6b328dbe02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 23:19:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2487213acea4485ba1b1d57a2d829cce\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/371f6cee30564f4a9f7864723c833148 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/291a5fa487824ab3a2e14f39ed33607e : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dcf05baa57524785b10d2ab903e17009 : SUCCESS in 15m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"89b64822f08619b7be9deb927b853b3aad8b430f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-23 23:22:42.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":9},{"id":"c468101c5f83bb0e8abef08fa1249ff4b78363a7","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-23 23:29:48.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/48189474fd914834a4d735f2b9716b87 : FAILURE in 23m 07s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":9},{"id":"6261ae369567beb5460e38f7caeadbb5f153e249","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-23 23:55:45.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-fc-driver build/3f8e814202434cf5a0f3ed603cf9a5b3 : SUCCESS in 37m 08s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":9},{"id":"01aaa5a5e41ad9e0b5b94fb47bd208f13b804c07","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-24 00:05:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/982731 : FAILURE in 39m 24s","accounts_in_message":[],"_revision_number":9},{"id":"52861010d9d42e6bc917a50c2a6cd4de70a88ddf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-24 00:44:08.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ebb2e5ec83024cfd978cc5d2124b82ee : SUCCESS in 43m 25s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":9},{"id":"629bb54ce18699b4ab98bea65cd73ee3fe712b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 00:48:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35c0289bab644d09b687842fb0ab027b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26374f69582642148ff0415ad944cc45 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bda479b6ad74baab201f2cd2473e797 : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4772ace5f6ef4314902acf272295a39c : SUCCESS in 10m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ead1805042c84e799030df066dbbbc2e : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8262eb535f0442519e1f853f4b111f07 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/1302582fc4444aa4b47688e05df37bd4 : SUCCESS in 1h 03m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a122335455f426fbf94dd05b6868889 : SUCCESS in 34m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a3d756c1a3814f669b4b11b8d5737c1c : SUCCESS in 2h 04m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dad5078480243e98ef719c58c33a39b : SUCCESS in 1h 02m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71adb62ddeba4153892d9b87e273687c : SUCCESS in 23m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a36646d743d34c34836f5e5f59160a72 : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/60ca995d5ece469fbf8bf59b1533ad38 : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61d5ca9b54ea4d7b97c7d450c5cfd1aa : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2fbfb6fb34c476b985b2a1155004762 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/662b1ddabc7c44db879538c414a1417b : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9eb38c9fb12e410dae8eaac7aa70083f : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa262c324cd0450cbd0b300e648e18d2 : SUCCESS in 4m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8fbbdbe3af0d40fc8d361aad25ee6fe2 : SUCCESS in 1h 45m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ef77415c4160418389e896ef4f6dd5f1 : FAILURE in 1h 28m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/37b3828776dd4e46bccd9b3383737ff0 : SUCCESS in 1h 49m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e93f92002f7e45f583ae277a725d68b8 : SUCCESS in 2h 00m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e75199c06df74022832e573cd5dce101 : SUCCESS in 37m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/079fffffedd74542ad7fa3643f4f66f4 : SUCCESS in 51m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c07846dffa045b3a16b928ee4af4bf2 : SUCCESS in 1h 32m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1a6e5c2281ce4fef889f12ffc541af22 : SUCCESS in 46m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7669c798797f44c580509d61003afbab : SUCCESS in 1h 05m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/024bf1c882ac4f978036e9586b0dfc21 : SUCCESS in 1h 55m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92333368fba740788284fa576ac2ca83 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":9},{"id":"291c1463ec0e7f4b17ec3b7074bc38cab3efd178","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-04-24 01:00:03.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_04_2026_00_53_57\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"7fd8072d11c66928d130dc27d2c8e5bbc6d444e1","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-04-24 02:49:11.000000000","message":"Patch Set 9:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982731 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"972ef06838b73ff9335f9eaee130e36782810c6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-24 07:27:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3943/.","accounts_in_message":[],"_revision_number":9},{"id":"0b1aec1ea9cd9a417c9d0b8a3f40f48f50d3b367","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-24 07:40:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3767/.","accounts_in_message":[],"_revision_number":9},{"id":"5e21a82b3652ba9080d9397d599f4145cfa2bb65","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-04-25 17:50:27.000000000","message":"Patch Set 9:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982731 : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"ab564600c8c8f784d6bc97c9e0543f375dc0e4d0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-04-25 20:41:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-9-fc : SUCCESS in 1h 5m 51s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-9-iscsi : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":9},{"id":"584090a94fd4ba46670a18606de12b22d730a025","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-05-15 14:51:54.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c3210ff9ead7d237f6752a4afb82093f063c676c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-05-15 15:56:55.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ffede0bdbd990e63059a504d68201e10b2edff90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-05-15 18:49:15.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"7c612fd69c1af2aa6e21da8d275df8f23323b139","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-05-15 18:53:28.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a969e21c59f612e3dfe0c624484e1f151344ba42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-05-15 18:54:24.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"b1b0f7733bef70070c695dde6c3fa8b10735392f","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-15 19:29:03.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hpe-fc-driver build/3971693be9e84cc19b0ee2dcc50d1982 : FAILURE in 33m 17s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":11},{"id":"77c62834d94741998f0598cd2b9c4418d8725424","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-15 19:36:14.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/982731 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"214cfac59623e0c6116757d51664a87301ca64ce","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-15 19:42:03.000000000","message":"Patch Set 11:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/982731 : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"03865492a1fe4023ca8d79ecddd23dddeb4533ef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-15 19:44:31.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":10},{"id":"17ae12a50b625a70b42da3eff2af695d3bc53909","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-15 19:52:02.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":11},{"id":"656c9c86eb9ef692d52b6ad14aa900c222069bc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 20:17:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e7034c1fac4600aa63554cbbffb263\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4de02845d6144885891d8b5b7a56a173 : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c39768e930fe4f26ac08d32c3959e6a0 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ebe79fae6fad4902b3384338003247c3 : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6df2d745b1069e22b119855a9bf453f32bb4b8de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 22:17:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3986/.","accounts_in_message":[],"_revision_number":10},{"id":"5ac73757be627cada6b9bc0fd4a6bdd7817d6a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 22:33:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a3783e5e94d413193dce52b6cc0444c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7701b8f186454f5c88b99870c66278af : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d91928d17084e8faff9728d265c0e8e : SUCCESS in 12m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51aba9ca0f5e407280b3a7285635723c : SUCCESS in 10m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5944984cf3de400889e88eefa0e9905f : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b98796608b34c76b7ee9b97994cf984 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/bad91206258f489a972517a2256ea55c : SUCCESS in 1h 06m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f82ac4041e3c494b8d0fd2c670557bcb : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5dd89323cec4ac4b4f26e74853b87d1 : SUCCESS in 1h 40m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14432db287044788b68f10400d05cb01 : SUCCESS in 1h 09m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fe8ed297a024d72b96f693164938389 : SUCCESS in 10m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/abe7d3f4832146408005eb38fad6fd3f : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5426d7d0179d4092a1c148f43ba1a639 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d33cb3d332c64e2ea1ae84ad3974db7e : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/24a1bc9b138d4ac1898aacb17c4b2ad1 : SUCCESS in 9m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6e2fa90aa5b34ecba5a48f04f27ddaa1 : SUCCESS in 5m 20s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/fc09d2d76abc4d5cadd879810f5bb154 : SUCCESS in 12m 18s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b78cc92870e44ffa6a25a368b918bf1 : SUCCESS in 42m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8d3e3ba9e0e418eb580c0dd0d4b98fa : SUCCESS in 3m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3fd42fc351a42938d8d36990c272999 : SUCCESS in 1h 46m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3bf1d31b8dbf41689d7657ca388a0670 : TIMED_OUT in 3h 25m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3decf2d4a4754653b804b3a9290935f4 : SUCCESS in 1h 53m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/89815cee891346388f58f017e9a52e09 : SUCCESS in 2h 24m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/310ba02fac47446d9f076ad43b88cc1c : SUCCESS in 38m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad1af1802ceb4f8398bb6f08709b7480 : SUCCESS in 1h 18m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ecabff69270f4b8d88c2b2b8816d1a5b : SUCCESS in 1h 10m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7371b33630646bc89f23b2aa71e4801 : SUCCESS in 49m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2af9078eb8334323979876afb2e1daa6 : SUCCESS in 1h 21m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19414f9732fd49039181a5c64ece6ad5 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":11},{"id":"13967e2c3a25064d7e90423981ecc75ccbb328e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-15 22:43:04.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_05_2026_22_36_27\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"2349f948f581e5c6b3523c16b99e715ba145e6e0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 23:22:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3987/.","accounts_in_message":[],"_revision_number":11},{"id":"c4157277b341a2e6927ff17446b2539a745b3134","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-16 00:51:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4162/.","accounts_in_message":[],"_revision_number":10},{"id":"d8d2388315c82163b7d2dd38c08c4e1b5dcb9651","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-16 02:04:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4163/.","accounts_in_message":[],"_revision_number":11},{"id":"5e0eb7ca307081657ad37cd24ea815d81e2963fd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-17 09:17:40.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/64b71c765a7b4ef2a063cd557432a693 : FAILURE in 28m 36s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/982731/etc/tempest_log.log","accounts_in_message":[],"_revision_number":11},{"id":"1fe021d446c0074c093c2008d39982b6d8b72a91","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-25 14:58:03.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-11-fc : SUCCESS in 1h 15m 38s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-11-iscsi : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":11},{"id":"7f868d6ddabaa7c951aaa2a7f8cf96a1b810faf5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:37.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5},{"id":"dcba9c108878f956b4d9b2d9a73cdaf7c4e80c7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:47.000000000","message":"Patch Set 8:\n\nBuild Failed","accounts_in_message":[],"_revision_number":8},{"id":"764e66701c964849be3a52940f55479e932d041f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:58.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"a87c176cd2c99a4a9600dc602a0b1f420f9e97db","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:34.000000000","message":"Patch Set 6:\n\nBuild Failed","accounts_in_message":[],"_revision_number":6},{"id":"b999b0788f1b87bf52c565e49d857d75dc982b3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:36.000000000","message":"Patch Set 9:\n\nBuild Failed","accounts_in_message":[],"_revision_number":9},{"id":"18d576838eb6ab21ad0ef74e246d3d30745eef29","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:41.000000000","message":"Patch Set 7:\n\nBuild Failed","accounts_in_message":[],"_revision_number":7},{"id":"7b44526ae543eda30d556482e06dfc6dbbacdffc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:03:05.000000000","message":"Patch Set 10:\n\nBuild Failed","accounts_in_message":[],"_revision_number":10},{"id":"b7431bc67bae78294b325f3ac596f0ea9a3f0214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:55.000000000","message":"Patch Set 11:\n\nBuild Failed","accounts_in_message":[],"_revision_number":11},{"id":"4cb5df11739df1af648a008337270bd679b79083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2026-07-31 16:48:40.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"32d52ae086db52a862ab21c8943879ae68642826","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 17:13:01.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e6656e6281564975a8a457d07594d757/","accounts_in_message":[],"_revision_number":12},{"id":"9557b1b0ca86cbfe506174681604c30b00dd61b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 17:14:17.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4809/.","accounts_in_message":[],"_revision_number":12},{"id":"f18443d813acc7322070dbf2a43d6c8ed2544555","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 17:14:58.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4985/.","accounts_in_message":[],"_revision_number":12},{"id":"a66f0798453339e76efb192dd776edb462196e66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 17:22:34.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b6d8bd45494ff3982410cd459a8a01\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fdc2afa5a4f2424385c67ed225909e02 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01bc649a85794481beebf31ebb3778be : SUCCESS in 20m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c2be1b0300d431db6b346db57e6af98 : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3ab523dbbfa4ef81b48784bf618a176d31f26725","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 17:36:02.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e6656e6281564975a8a457d07594d757\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7e7ac7004a94aaca9480fbe086a2b00 : SUCCESS in 41m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b86ce80bc9141b798cf0c955ad52175 : SUCCESS in 37m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f788db828084321b2ce4e2da5832d5e : SUCCESS in 44m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/608def8cefc64951b391393d19183ad4 : SUCCESS in 39m 03s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/82fdf00fca374f79bb1f1727b696f420 : FAILURE in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"786bed64fb0a78e95c1bdfe12aeaf4c7251ff965","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 17:39:42.000000000","message":"Patch Set 12:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/683 : SUCCESS in 0h 49m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/982731","accounts_in_message":[],"_revision_number":12},{"id":"c1bb8bb5d65d8221aa1465913cea0684b80cfba6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 17:40:36.000000000","message":"Patch Set 12:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1044 : SUCCESS in 0h 50m 44s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/982731","accounts_in_message":[],"_revision_number":12},{"id":"1f4fd942a1e86346565c77b907848134c7ff76be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-31 17:45:57.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/18ce6083656048b1a4c5c8c6d70916b1 : SUCCESS in 46m 15s\n- cinder-storpool-tempest-iscsi-multipath build/c306ff8acdd0415cb09360eb06e871e9 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":12},{"id":"5dea65417611bf131e17f1eff8e97a490df33207","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 17:46:27.000000000","message":"Patch Set 12:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/774 : SUCCESS in 0h 49m 47s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/982731","accounts_in_message":[],"_revision_number":12},{"id":"e52f9e28cfe9235711ae9700a65274a8fe93c71f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 17:58:54.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":12},{"id":"75dd911a3d8af2a7b51c1c96c26b985b52237731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:32:31.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/902a2c339deb4c71a6334f58eee36c36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae855fc27c19436da498efd6328ac969 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24cbc559f2ea4cfcabf18718a5914ede : SUCCESS in 11m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0d5c9175e804008af1c57ecc04c9592 : SUCCESS in 9m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dfd21d62900c4042b0e73382a6699e18 : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c1e2eb403749e49e4061de94d17ab0 : SUCCESS in 14m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/39af4d69d81c4766887d74db1a5d28ff : SUCCESS in 59m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d7f8a5efe2f4bda861431c0affd16fe : SUCCESS in 39m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5886bf28ff704911a1a5c5694b12e7c8 : SUCCESS in 1h 23m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf278fd6f3a94604876fa5c658133756 : SUCCESS in 1h 00m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bf36eee79a34188b067972264795ed8 : SUCCESS in 10m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0dc749e95db643f6aa2aa20b67732af2 : FAILURE in 27m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64f8d6b491bf475eb27be531b0b19dc0 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4b26995a4404415c891bd1641a23b717 : SUCCESS in 7m 02s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/d460da6a7c66481cae9e6e8cda67e75e : FAILURE in 13m 25s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/61c2f77c8d6a40ff9dccb5a067e041b6 : SUCCESS in 6m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/946911d67fa14873bb6ce7218ce5a1e4 : SUCCESS in 7m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6f1aa134e23843fda8eb93115f483b17 : SUCCESS in 8m 22s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/234b083a25654d61a689932e0a47979f : SUCCESS in 38m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/79a71b36b4414264acdab744b83e6098 : SUCCESS in 3m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/da0c8ae87c964af78965920ac1614bd5 : SUCCESS in 1h 01m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/50a525cca5864cbbade9b8c43f3199c3 : FAILURE in 1h 51m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b87517d17e2b48fea612260a61f53c27 : SUCCESS in 1h 46m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/056168f8fd26496d9b4f0933c79afa99 : FAILURE in 2h 37m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/aa03ed9bcfe841a7bec0bd865f086e94 : FAILURE in 1h 37m 20s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/880d2fe59673474a82cdbedc4227036d : SUCCESS in 38m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/984650de90494d0eb5f0d09441763e93 : SUCCESS in 52m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1f5fe5789e14ec49c22661799375b9f : SUCCESS in 1h 19m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/95f2ccfc9ef9430f9204dd80d4ab35ed : SUCCESS in 50m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/11f87a9b45064a5c990ea61acd299a4c : SUCCESS in 48m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/139472bc67d44cad82c8678581f97c05 : SUCCESS in 53m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"edd4f59ac4300fe1d51a8f79b14a81f448080e1a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 19:46:18.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_07_2026_19_39_34\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"87ac5ae9ab202b85e70b7ac58320bf10542cd437","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-08-02 04:22:54.000000000","message":"Patch Set 12:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/982731/12/20260802082901/NEC-ISCSI : FAILURE in 4h 20m 34s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/31/982731/12/20260802082901/NEC-FC : FAILURE in 4h 53m 49s","accounts_in_message":[],"_revision_number":12},{"id":"382e0c4b64f74dc964c9824406bcd2d897660628","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-21 16:51:01.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"831c1f958c1f253f5a4c13a37049ac2ecacf975e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 20:15:06.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9a2bac97beebc506db91fe97929c380f79d03c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 20:16:16.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"e9812179423141b4a4192f60c2c81d26a142f509","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 20:16:46.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2f42cacc3545f5b837a39de7ba65e9d2ebef8a6e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 20:24:09.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5240/.","accounts_in_message":[],"_revision_number":13},{"id":"c2fb946169b9486f630f929d77dc1871382c7cf0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 20:32:43.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5241/.","accounts_in_message":[],"_revision_number":14},{"id":"69b4e43706a8a78fa7d01702bf323dca7933a1f2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 20:53:39.000000000","message":"Patch Set 14:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/23160e81b499465eb25f579434609c7e/","accounts_in_message":[],"_revision_number":14},{"id":"44435699cb8983d37f04b7536633921cdd5fd4a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 21:11:49.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":13},{"id":"ebb3db7c8444a2fc541e4781b63b486fd4430475","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 21:13:52.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":14},{"id":"a8981d35140dcf3ccc3eb25b42ff055f00752f41","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 21:25:26.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/23160e81b499465eb25f579434609c7e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9e383bb924d448d7b2c9f0a13a1fc5ab : SUCCESS in 34m 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a0b65a14ecc4155b9c6f027d087032c : SUCCESS in 36m 59s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7fde9c8105cf4752827cc4c947303dfd : SUCCESS in 40m 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41eb27f5d9c44ad489da49abc3aff25e : SUCCESS in 31m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab89d3105cd840c587c8844f38f4f90a : FAILURE in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8a37d757ed860dcd2ab5be826f45f200ed4e007f","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-24 21:39:06.000000000","message":"Patch Set 14:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/927 : SUCCESS in 0h 46m 31s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/982731","accounts_in_message":[],"_revision_number":14},{"id":"20cbee6518b5c6a88c9ed69814a64e2d2c5f904b","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-24 22:14:34.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2594f8c49963fc0945712f4c447120bcda1b38b7","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-24 22:16:36.000000000","message":"Patch Set 14:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1020 : SUCCESS in 0h 47m 7s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/982731","accounts_in_message":[],"_revision_number":14},{"id":"1c0d19999ca33a32d1e89ee575dbe116d9c6c99c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-24 22:18:31.000000000","message":"Patch Set 14:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1289 : SUCCESS in 0h 47m 9s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/982731","accounts_in_message":[],"_revision_number":14},{"id":"65df3caebb597028f7400829df115126979fcf45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:34:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d74abcbd915444839f94913df6092b7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e5d7aa8a9b49b8bdd01cad9af75db8 : FAILURE in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6514589037ab4c9ba7dc795c07345aca : SUCCESS in 9m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/562d02245c5344bd8ef572c21f250b82 : SUCCESS in 9m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a141fba993744e2afe047f9cc8190e1 : TIMED_OUT in 41m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07a9db3f73843c2ba80ef84bdf58b00 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/ba0ea9fd86eb43a682958781aac4ffbd : SUCCESS in 1h 05m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf9737fb53eb493fbc0dff0e231af5e9 : SUCCESS in 1h 03m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2aa4303ae6f49029378a21a1d502346 : SUCCESS in 59m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/212f9100a9e64faab381b2dad98bdf3a : SUCCESS in 33m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75371b88b30b4e599259b230e87e4876 : SUCCESS in 9m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a998d65ff01944e5921c316e25f14f25 : TIMED_OUT in 41m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fc36d357296e4f99955c87c7b4be2e2d : SUCCESS in 4m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/038f2f68ef474b7bbebbf28a65191cb6 : SUCCESS in 6m 09s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e9f54a1eee8f4f019b064b5a97789a62 : SUCCESS in 10m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/db183c3b4b334341840e78705006ce2a : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8677a77598b84a05b53550d6921dd004 : SUCCESS in 6m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8a06532d35f14a9cbff3837bc6bbf45a : SUCCESS in 5m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5b27f514d7614555b8889941c2dd138d : FAILURE in 6m 55s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8258ba99f4794510b0e2deaa45f8969f : SUCCESS in 8m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/781eac135dfa4136b970225293f2348e : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cdd7719f809641fb8716e6dadda2857e : SUCCESS in 3m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7479d921c23844ea91bc0086959ce93e : SUCCESS in 1h 39m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d930c56741aa4aefa4c2c2e9672d8380 : FAILURE in 2h 06m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb685c23eb674295b118f7a7ebc53026 : SUCCESS in 1h 58m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/81626639bea14318b0da24dcc7f3f32a : FAILURE in 2h 13m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/61e39f78a0804e3f8ffdd4029b04fbbb : SUCCESS in 2h 03m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ba907914c0f2409db213656ef51ab4a3 : SUCCESS in 37m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c612d49223b4472a826e020b44439894 : SUCCESS in 1h 20m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3f691c6bea2d47179948b0794280083c : SUCCESS in 1h 41m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f3c7f9979a348b69754b7b189c89e09 : SUCCESS in 42m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/21f2d5a948c24e28b8766eb1c5d21eca : SUCCESS in 1h 40m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71e7c1ce35924bc3bf7d123d66b496b7 : SUCCESS in 30m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"1f7428eba0ce194ba8d678b8bb415cef27a6b3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 22:47:08.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"6691653c4af725328ad65a0a5fe234214525c9f2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 22:56:23.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5242/.","accounts_in_message":[],"_revision_number":15},{"id":"17188f6187c849b8de5d21231e2ae0d1bf9a40e6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 23:17:06.000000000","message":"Patch Set 15:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1361d38fa3094ba98c0a7f04e26acbe6/","accounts_in_message":[],"_revision_number":15},{"id":"7e69ccd28375ac44ab827e2257f4fe1b8feb5767","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-24 23:18:00.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"fe40dc23fdfe75c77858d87f924be8e7d647484d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-24 23:36:19.000000000","message":"Patch Set 15:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1290 : SUCCESS in 0h 48m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/982731","accounts_in_message":[],"_revision_number":15},{"id":"237bc8cbd7c8cc854f7cfbd19e5abaa3fb82e402","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-24 23:36:25.000000000","message":"Patch Set 15:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1021 : SUCCESS in 0h 48m 52s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/982731","accounts_in_message":[],"_revision_number":15},{"id":"560ef50c44f1116920a09e9700d62ccb7a292b52","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-24 23:36:53.000000000","message":"Patch Set 15:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/928 : SUCCESS in 0h 49m 23s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/982731","accounts_in_message":[],"_revision_number":15},{"id":"9c708bf2e80755c2bb4a15b0887850d53da6a0e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-24 23:39:59.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ee88938f5e6d4d4da43c9da803e411e4 : SUCCESS in 47m 34s\n- cinder-storpool-tempest-iscsi-multipath build/9ecc51bd141b40e79d92352d59fa7350 : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":15},{"id":"baad8c1fa7270e4f8558c31d6adec633b795f226","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 23:41:59.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":15},{"id":"3560e8f5b8af8f88b9ea437d5003774ad3ad3eee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 00:20:06.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1361d38fa3094ba98c0a7f04e26acbe6\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c4930ab80204f8fa3c97fea267b0074 : SUCCESS in 39m 02s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0295cc86ece049798ad6a690050e980a : SUCCESS in 29m 31s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0ddabd8597046b2a294115b71523222 : SUCCESS in 42m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d64c79a6ee7411b9839ee3d10fa8e75 : SUCCESS in 32m 19s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d099b539f9d54be18ab1277684cf169d : NODE_FAILURE Node request 200-0000968030 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c9e14c4478f348f665fd5b2203cc94a3120fd2f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:22:38.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2344b086ee404590abcdbaf11fed6049\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91708c4cd7e74076a1fa4f53420d211d : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c2a7999fef7e499c8cef6f0f9a474782 : NODE_FAILURE Node(set) request b6e9e9dedfa7474eb7b9e314ab2efd76 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b743c290e6f44049fa2f88b32daf0c5 : SUCCESS in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0672358be4957644fbe881cb1d7a43bf8913e10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 01:41:00.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb0c7031d874a5ea78b68e264db43bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa64df781f6f44489f090a3a75e51757 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4566cab8e5fc48239183820dca41c678 : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/877419538abf40aaba302ec04bbc7157 : SUCCESS in 9m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/710401def31e4b9a9de1110a8263fb86 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144a06d137ad4e70a35ae9f59c69d194 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/218312cb0cfc437c9cb742e662298733 : SUCCESS in 1h 04m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08f424bd0b6b431f8dd1b8bfd0ed101a : SUCCESS in 35m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88d0266146c44b3c86ef2cbefee2ad68 : SUCCESS in 1h 07m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8d81e3d94c94a36ae88a3f1ba6098ba : SUCCESS in 1h 01m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0888111285a7495db52c73ec05b918e3 : SUCCESS in 6m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5b9d7c33c8a14bb79e56bf9c8b0b6e8a : SUCCESS in 20m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9f6737f061fe4ced9f267a9f28f59950 : SUCCESS in 3m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2b5105f2683437e858a9f19afcae3df : SUCCESS in 4m 08s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/89a44808c96240249c9200003a0a52f5 : SUCCESS in 8m 36s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/25a65fa708564dadb9f23a0d7519d349 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ac075a5874384549b53c8c9b04a8bb08 : SUCCESS in 5m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/88af5c13add846dea6fc2cb06190038d : SUCCESS in 4m 47s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a2cf39a0fc014a77891a06ed82358de0 : FAILURE in 5m 06s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/45f33d6277aa443194a2e4d11e6e9080 : SUCCESS in 7m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89e42953d8e44afe84fe271e525d72f6 : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/377b94f8cda14b7f97869997b9342836 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c122f828362a4139a7b1250630bfbf6e : SUCCESS in 1h 34m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d5905d25a4794db8b3501a46ee4c614e : FAILURE in 1h 03m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e8b858e716764d688302b2a0082705a7 : SUCCESS in 1h 27m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/088d7c4f8bc6494cab584c4e46859878 : FAILURE in 2h 37m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1ce8b5ba9aab4bf6b8fee5429ff4ab4c : SUCCESS in 1h 14m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/97de05a1485c4b85921207f3494dfd56 : SUCCESS in 18m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3bfe67b3148472eb90de3ff11556967 : SUCCESS in 1h 07m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21ce41843c8148c8ab40398da9509423 : SUCCESS in 1h 35m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/341de208a3ab4f8191e87b376ebd742d : SUCCESS in 45m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e8e03c141694b93ab2df90f5883c64f : SUCCESS in 1h 33m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57d93b26b72f4a039f291896f9c9f414 : SUCCESS in 57m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"ee7cc4e8d53fc546bd184a1c85238d3f233e4638","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 01:51:36.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_08_2026_01_44_34\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"19c8f3b20e27a50ccbc298734805d35d791b2ab0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-27 08:58:11.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5064/.","accounts_in_message":[],"_revision_number":13},{"id":"7aadb9335f49bdc6d451391397ea546b41f88f3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-27 11:04:01.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5065/.","accounts_in_message":[],"_revision_number":14},{"id":"94f6c86693d791ae25effb7924ea39d229f87a35","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-27 13:05:58.000000000","message":"Patch Set 15:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5066/.","accounts_in_message":[],"_revision_number":15},{"id":"61af37058b5008dc03b604cb3ffda2c179003ecc","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-27 23:57:38.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"881bb356bb2d75167a12e8f238bec64846764ac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 23:58:30.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"783f7c8008f688336447057fe899380203ef7257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 02:54:41.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e22f73bac2ca43aa83ae419e3bcf4b43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2c724f17adb4f1aa63cccafa06dea76 : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc7b8fbdf34d4541830190089b7699ae : SUCCESS in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db81d7530ba548869892aa296e5a1bf9 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9000221dcf24704bba1367da25660ff : SUCCESS in 14m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/74ace205da644f69ad7f3ae990d56048 : SUCCESS in 29m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4031779c0094929beb14ceb4f923191 : SUCCESS in 1h 10m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c66f7be0059499098158aa526bef8b9 : SUCCESS in 56m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dad3dedea0f14bd78ac7168876c070db : SUCCESS in 46m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b60c66af9487441a9ceb190aa3029817 : SUCCESS in 9m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/caeda08e7ade47fbb058390e1afc2c7d : FAILURE in 1h 17m 32s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6023ee5521754a0693ed04a52783ec39 : SUCCESS in 50m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ba8a5bf0de34a3dbfe0ab5066e23cea : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":15},{"id":"9d1e25e64d8331ef3ab664d58facc8307b67fb49","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 04:54:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"806dc8810089d72190afa649b201d9af991cffba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 05:03:48.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nThe following files contain Git conflicts:\n* cinder/tests/unit/volume/drivers/hitachi/test_hitachi_hbsd_rest_fc.py\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"46792ebf6d80226d25134d67188b0cb4695daf6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 05:06:48.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"0cc930388536b9181f1f24a93ce2276fc78125a5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 05:06:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":17},{"id":"5b716fca846f4fa0743c9a7098f04f72f5b471da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 05:15:54.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5280/.","accounts_in_message":[],"_revision_number":16},{"id":"5f7bb56b1fbe63cc774dacf139f6c6625c3b7044","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 05:34:33.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5282/.","accounts_in_message":[],"_revision_number":17},{"id":"710dce8b8c9399212e2009de0020c9508feb3d2c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 06:06:54.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":16},{"id":"4e39bad984c1ffed4a1138e15b9d0e2f3b1cdfa6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 06:10:59.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I20267d3011e2c7de1cfa18e229fb545f4394e48b .","accounts_in_message":[],"_revision_number":17},{"id":"ce64f43dd6ebc98b380a962c22131a0007950566","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 06:15:00.000000000","message":"Patch Set 17:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b8852030f8f64c90a3eb18978d7e345d/","accounts_in_message":[],"_revision_number":17},{"id":"ad53468601d66099281ba535fe8a8f104190897b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 06:26:51.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/589f1939190246289a0003c7767a4cf0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/72eadac6e6e94d898d787a05263c3cae : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ecae6a68caeb4675aa4f3fb7a64a339a : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c0e90f7a48445f98bd9abfb1dd74998 : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3683c39e07a989ff811ce0a77dfa49c8dc383879","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 07:12:20.000000000","message":"Patch Set 17:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b8852030f8f64c90a3eb18978d7e345d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/aa6be538d478428bbe7316af6079b84c : SUCCESS in 39m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb88368dedd647aea7a10e1bbbc07888 : SUCCESS in 38m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c37d1c4fed948378d06a9f897e6b68f : SUCCESS in 54m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22ec924802d14c0aae68bb91020a2ba6 : SUCCESS in 37m 04s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f067ab95d6d04166a215b97a08da00d3 : FAILURE in 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"28edaba6d395d976570b6b62c5416d91f152dd0e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 07:30:57.000000000","message":"Patch Set 17:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1330 : SUCCESS in 0h 46m 51s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/982731","accounts_in_message":[],"_revision_number":17},{"id":"9dac74cef63bfa19b276a8787a467982345b981e","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 07:32:03.000000000","message":"Patch Set 17:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1061 : SUCCESS in 0h 46m 59s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/982731","accounts_in_message":[],"_revision_number":17},{"id":"9e90b91c10fcaecf017bb0fe09939c29998550bf","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 07:32:17.000000000","message":"Patch Set 17:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/968 : SUCCESS in 0h 46m 37s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/982731","accounts_in_message":[],"_revision_number":17},{"id":"7343de57931304ffedc7b9efe7676529cab4b9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:50:00.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff31bca19b1a47adbd97be74942bafb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1434b146857d4051b38d0b8344d8fe71 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d25b919dedd94f56bd3a4da5cc280de8 : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba68d0ec37924572904a3d95f1f53e58 : SUCCESS in 10m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d6513520c7e845c78e01e859d1feb74d : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0496c81e725148c2b5cbe5d96ef10983 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/3fc7142d6a0443a9ac74e953ec4f5c6f : SUCCESS in 1h 03m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74a0683a2267489b833a58ff032f0395 : SUCCESS in 53m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d65ca8a1b0924d739d314d2204a82335 : SUCCESS in 1h 28m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ac38971756247a69c04ff24db92b7ba : SUCCESS in 51m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cbd96d7969c4f60acfd4360349914db : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b767b9057fd243f78a4906918b5d80d5 : SUCCESS in 18m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f837e9d79c0e490aa5dc4c3e6e35a625 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1900e4b9e7b243f1b142a3443fc2a5da : SUCCESS in 6m 10s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7d2f59b7e7cf49529f15f06192afbbc8 : FAILURE in 11m 19s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/39b816fa1f41483ca3797f4f4e8d4296 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b0bf090c76d94de58179847143e62e00 : SUCCESS in 6m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2162c5a62cdd4c5faa238ed62b9d6035 : SUCCESS in 7m 02s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f6c2977175bc45d6aac0ce1212f7a46e : FAILURE in 7m 25s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8add5b2c69c542e5a75102e861db22a6 : SUCCESS in 8m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0db59863f5a340d1ae7ca7546c728b15 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/307003741539489497986b25bba27c1b : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bbae70940b6469abcd08b36ab046b9d : SUCCESS in 1h 47m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/869e66686f4844548448f4a67783bee2 : SUCCESS in 2h 14m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65cea6774c05485187574bd581246f5c : SUCCESS in 1h 31m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/55d872902d2f41a98f383c5261fa0477 : POST_FAILURE in 2h 41m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/671a98be320e462998a388b80b01b7f9 : SUCCESS in 1h 37m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6b8247c986f3406db1579077f6ce9194 : SUCCESS in 38m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e19e28ee9070475ebddc8e1bf03f88fe : SUCCESS in 40m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8cc0de4abf9f418fb800bd38f4a70f28 : SUCCESS in 1h 12m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e3fd18ae57e44011971b52f978c205be : SUCCESS in 41m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b26dc2dd557b48edb4bd55b4123a36e8 : SUCCESS in 40m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbfa695ccd85425a877d00aae9fdeabe : SUCCESS in 40m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"182d551bed61acf21014255170cd98090c251e99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 08:01:11.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/810c899816a04159acbbe1617b8413f8 : SUCCESS in 43m 08s\n- cinder-storpool-tempest-iscsi-multipath build/e4d945b1882f43cab93e128a189492b8 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":17},{"id":"ef14284274c0423c22aff35266c2c5fb6c80083d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 08:05:43.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_08_2026_07_58_46\nLeave a comment with","accounts_in_message":[],"_revision_number":17},{"id":"4d87de747d3cddbac255b78123ec615a04f64c43","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 13:15:52.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-15-fc : FAILURE in 9m 52s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-15-iscsi : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":15},{"id":"60d899dea0276149a4848cbf8b375310d54dd8a1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 17:06:10.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":17},{"id":"34b27b71ef106b1cba4dd295f0f2acb02b0f4b54","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-28 17:35:19.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"a53e40a217f95cf22d7d669cd33c524049cc3630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:35:56.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"a9e66bbc4de4916c6f9811099f973951b5e77899","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-28 18:08:28.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"79be3975949c5b977d983b66f6742bc2622dec2f","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 18:37:49.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"fde7676d5b5c888e6d628520b9a4369c60e53ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 19:26:55.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5444407f49643ef96fd1bb673ec97d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b6116e55d3421988bb8d7416455069 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/042ecb315a6549ba83bb4a8d85103645 : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f3d433bafe3461f94c6e2373b9fb181 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7f05e73b014f1f96bb7b057a3506ab : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/cfe78cc687b44d528bd8997ae5141f9e : SUCCESS in 1h 05m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74765fbfd65e47c7883948624db529ff : SUCCESS in 1h 06m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5eaf018f6fad42b6b3a415b9cd806306 : SUCCESS in 1h 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05ef5144cd71476ba7eef90958dd2720 : SUCCESS in 31m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/722d10ffa3bb4c5a8f82f73c90038922 : SUCCESS in 10m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3864189589e4b07b374724c458abdc2 : SUCCESS in 1h 09m 17s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b529508cde5f4deb800d3ca96c9c07c1 : SUCCESS in 1h 29m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d55cc5c57853459faf5ccb3cd7cedf5b : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":17},{"id":"791877cea4e42ec2eaa83924be188fbd2e65492f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 19:26:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"6fa4b4349d67dbdbe2e083acd67bd86a80079627","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 19:28:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d1be0f33aa4fe5bf7697401d87ef47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e69fb3383ec485e97697ce8c0b0ffed : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01c5204698344f987c643b2bd721203 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":17},{"id":"022217cd0e291fb08389dda2aedfe886b76a5ca1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 20:06:48.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-fc : SUCCESS in 57m 20s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-iscsi : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":17},{"id":"c7b6a9296dad3afb74f8a3365341513ac75e3210","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 20:44:22.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-31-982731-17-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"2b2b200ba784e91912328ca03e171f90e8fae8d9","revisions":{"40d4fd6b1ff555ec35d29c227ad8317eda0ee72a":{"kind":"REWORK","_number":1,"created":"2026-03-30 22:36:56.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/1"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:37:17.000000000","tz":-420},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40d4fd6b1ff555ec35d29c227ad8317eda0ee72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40d4fd6b1ff555ec35d29c227ad8317eda0ee72a"}]},"branch":"refs/heads/master"},"f782e62b67c6734f02c139a6c4ea8490c12123e8":{"kind":"REWORK","_number":2,"created":"2026-03-31 03:28:15.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/2"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-31 03:28:50.000000000","tz":-420},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f782e62b67c6734f02c139a6c4ea8490c12123e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f782e62b67c6734f02c139a6c4ea8490c12123e8"}]},"branch":"refs/heads/master"},"892d9a1b15dccbda90b8b63dfc50962997ef06f6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-17 23:19:06.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/3"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-17 23:19:06.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size (Closes-Bug: https://bugs.launchpad.net/cinder/+bug/2148722)","message":"[Hitachi] Fix B20 snapshot creation failure due to different size (Closes-Bug: https://bugs.launchpad.net/cinder/+bug/2148722)\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/892d9a1b15dccbda90b8b63dfc50962997ef06f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/892d9a1b15dccbda90b8b63dfc50962997ef06f6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5c09d55974d97d3ba9e653e8855437ffb6e0262b":{"kind":"REWORK","_number":4,"created":"2026-04-20 16:52:28.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/4"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-20 16:52:28.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size (Closes-Bug: https://bugs.launchpad.net/cinder/+bug/2148722)","message":"[Hitachi] Fix B20 snapshot creation failure due to different size (Closes-Bug: https://bugs.launchpad.net/cinder/+bug/2148722)\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c09d55974d97d3ba9e653e8855437ffb6e0262b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c09d55974d97d3ba9e653e8855437ffb6e0262b"}]},"branch":"refs/heads/master"},"79d97252f056e15f59ffd2f1b94530eca9b41ea7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-20 17:18:10.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/5"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-20 17:18:10.000000000","tz":0},"subject":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source","message":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source\n\nCloses-Bug: #2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79d97252f056e15f59ffd2f1b94530eca9b41ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79d97252f056e15f59ffd2f1b94530eca9b41ea7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e756a039d9d434dbc6489c1a43bf2d8a2a225270":{"kind":"REWORK","_number":6,"created":"2026-04-20 23:12:03.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/6"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-20 23:12:03.000000000","tz":0},"subject":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source","message":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source\n\nCloses-Bug: #2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e756a039d9d434dbc6489c1a43bf2d8a2a225270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e756a039d9d434dbc6489c1a43bf2d8a2a225270"}]},"branch":"refs/heads/master"},"404e9d9de50656df66abee5f6b78d93ea74b5302":{"kind":"REWORK","_number":7,"created":"2026-04-21 04:06:28.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/7"}}},"commit":{"parents":[{"commit":"4c92ca90b6b82a06fc852711911e6a1f742a1563","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c92ca90b6b82a06fc852711911e6a1f742a1563"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-04-21 04:06:28.000000000","tz":0},"subject":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source","message":"[Hitachi] B20: Volume Creation from Source Fails when Size is Larger than Source\n\nCloses-Bug: #2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/404e9d9de50656df66abee5f6b78d93ea74b5302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/404e9d9de50656df66abee5f6b78d93ea74b5302"}]},"branch":"refs/heads/master"},"dfc08e1044b67640215a8a478970351f429180a0":{"kind":"REWORK","_number":8,"created":"2026-04-23 17:26:26.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/8"}}},"commit":{"parents":[{"commit":"fb7f818a87bfdf62d643f3136464e2d358483390","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb7f818a87bfdf62d643f3136464e2d358483390"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-23 17:23:25.000000000","tz":-420},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfc08e1044b67640215a8a478970351f429180a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfc08e1044b67640215a8a478970351f429180a0"}]},"branch":"refs/heads/master"},"96c36642551af59591b0692b73616ee4441a9cfb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-04-23 22:38:11.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/9"}}},"commit":{"parents":[{"commit":"96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-04-23 22:38:11.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96c36642551af59591b0692b73616ee4441a9cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96c36642551af59591b0692b73616ee4441a9cfb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"dfc08e1044b67640215a8a478970351f429180a0","theirs":"96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc","contains_conflicts":false}},"c5ca4f38ceb9e5b0abd1830f137a17702d66e6e7":{"kind":"REWORK","_number":10,"created":"2026-05-15 18:49:15.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/10"}}},"commit":{"parents":[{"commit":"96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-05-15 18:47:35.000000000","tz":-420},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nFix bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5ca4f38ceb9e5b0abd1830f137a17702d66e6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5ca4f38ceb9e5b0abd1830f137a17702d66e6e7"}]},"branch":"refs/heads/master"},"8a5b6877d3f2efbde2c08878610aae9b817b95c5":{"kind":"REWORK","_number":11,"created":"2026-05-15 18:54:24.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/31/982731/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/11"}}},"commit":{"parents":[{"commit":"96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc","subject":"[Hitachi] Debug Enhancement: Introduce instrumentation for auditing http/s requests to assist with debugging (Driver, UT, or Tempest).","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/96a781b4c47cfdcc3d1a0d110089c5fb8c93cbcc"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-05-15 18:54:24.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nFix bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a5b6877d3f2efbde2c08878610aae9b817b95c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a5b6877d3f2efbde2c08878610aae9b817b95c5"}]},"branch":"refs/heads/master"},"9f06463f323ef9f9e6b4cf65f59470b9e378c024":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-31 16:48:40.000000000","uploader":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"ref":"refs/changes/31/982731/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/12"}}},"commit":{"parents":[{"commit":"7d9d73f18d527c1ec25266952819d611011a5227","subject":"Merge \"doc: Add bug reporting template for community feedback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d9d73f18d527c1ec25266952819d611011a5227"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"anthonygamboa","email":"anthony.gamboa@hitachivantara.com","date":"2026-07-31 16:46:54.000000000","tz":-420},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nFix bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f06463f323ef9f9e6b4cf65f59470b9e378c024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f06463f323ef9f9e6b4cf65f59470b9e378c024"}]},"branch":"refs/heads/master"},"40b86f7bf1651f4b928f49437e6fbf7a32f230d0":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2026-08-24 20:15:06.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/13"}}},"commit":{"parents":[{"commit":"7d9d73f18d527c1ec25266952819d611011a5227","subject":"Merge \"doc: Add bug reporting template for community feedback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d9d73f18d527c1ec25266952819d611011a5227"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-24 20:15:06.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nCloses-bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40b86f7bf1651f4b928f49437e6fbf7a32f230d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40b86f7bf1651f4b928f49437e6fbf7a32f230d0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9434ba92268c2cad3d2618272316aeeb77e438b2":{"kind":"REWORK","_number":14,"created":"2026-08-24 20:16:16.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/14"}}},"commit":{"parents":[{"commit":"7d9d73f18d527c1ec25266952819d611011a5227","subject":"Merge \"doc: Add bug reporting template for community feedback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d9d73f18d527c1ec25266952819d611011a5227"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-24 20:16:16.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nCloses-bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9434ba92268c2cad3d2618272316aeeb77e438b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9434ba92268c2cad3d2618272316aeeb77e438b2"}]},"branch":"refs/heads/master"},"924bf2a3a6961d52f9b480724ca8cd877e4c8aea":{"kind":"REWORK","_number":15,"created":"2026-08-24 22:47:08.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/15"}}},"commit":{"parents":[{"commit":"7d9d73f18d527c1ec25266952819d611011a5227","subject":"Merge \"doc: Add bug reporting template for community feedback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d9d73f18d527c1ec25266952819d611011a5227"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-24 22:47:08.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nCloses-bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/924bf2a3a6961d52f9b480724ca8cd877e4c8aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/924bf2a3a6961d52f9b480724ca8cd877e4c8aea"}]},"branch":"refs/heads/master"},"38ee949cfb42c6754cfb8f367e914de4981b101d":{"kind":"REWORK","_number":16,"created":"2026-08-28 05:03:48.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/16"}}},"commit":{"parents":[{"commit":"77c06c8ba681938376e612d55c7646891da96012","subject":"Merge \"[Hitachi] Fixes bug where reclaiming space does not occur automatically.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77c06c8ba681938376e612d55c7646891da96012"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 05:03:48.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nCloses-bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ee949cfb42c6754cfb8f367e914de4981b101d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38ee949cfb42c6754cfb8f367e914de4981b101d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7d9d73f18d527c1ec25266952819d611011a5227","ours":"924bf2a3a6961d52f9b480724ca8cd877e4c8aea","theirs":"77c06c8ba681938376e612d55c7646891da96012","merge_strategy":"recursive","contains_conflicts":true}},"2b2b200ba784e91912328ca03e171f90e8fae8d9":{"kind":"REWORK","_number":17,"created":"2026-08-28 05:06:48.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/31/982731/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/982731/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/982731/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/982731/17"}}},"commit":{"parents":[{"commit":"77c06c8ba681938376e612d55c7646891da96012","subject":"Merge \"[Hitachi] Fixes bug where reclaiming space does not occur automatically.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77c06c8ba681938376e612d55c7646891da96012"}]}],"author":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-03-30 22:35:56.000000000","tz":-420},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 05:06:48.000000000","tz":0},"subject":"[Hitachi] Fix B20 snapshot creation failure due to different size","message":"[Hitachi] Fix B20 snapshot creation failure due to different size\n\nCloses-bug: 2148722\n\nChange-Id: I20267d3011e2c7de1cfa18e229fb545f4394e48b\nSigned-off-by: Hao Chen \u003chao.chen1@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2b200ba784e91912328ca03e171f90e8fae8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2b2b200ba784e91912328ca03e171f90e8fae8d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
