)]}'
{"id":"openstack%2Fcinder~984410","triplet_id":"openstack%2Fcinder~master~I7d7fa80af2f910c7a6b804d323309b42fa2988d8","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"38319":{"account":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"last_update":"2026-04-15 11:19:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7d7fa80af2f910c7a6b804d323309b42fa2988d8","subject":"IBM-SVF: Fix force detach volume bug","status":"ABANDONED","created":"2026-04-13 15:14:17.000000000","updated":"2026-06-19 19:55:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f0f694a2f4ed94ce91f362bd38bd169750af4f62","_number":984410,"virtual_id_number":984410,"owner":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-13 19:47:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 15:20:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-13 15:52:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-13 16:24:06.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-04-13 16:24:27.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2026-04-13 19:47:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 21:47:09.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-04-13 23:19:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-14 04:11:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-06-19 19:55:30.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"1182a900822afded6f0393b0c25446b8e7cd979a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"date":"2026-04-13 15:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2a1b3d5716d22fc7a732a5a5b8b82718f452184","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-04-13 15:20:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9ea1cda109724c688f7676d690185eef : RETRY_LIMIT in 27s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/27e6bf1e1daf43cb975634e015ea9ec9 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":1},{"id":"218ea514a19a3d62683759cfa1fad16c4a688744","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-13 15:52:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I7d7fa80af2f910c7a6b804d323309b42fa2988d8 .","accounts_in_message":[],"_revision_number":1},{"id":"25244d39c03cd8de7ab003276912f04cf6a90060","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-04-13 16:24:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/7e2cd6ea96df49d0b0df819bf81e682a : FAILURE in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a6a9f1d90b9054a347891420a712d6bdb10f21f0","tag":"autogenerated:zuul:check","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2026-04-13 16:24:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/2789693f3a934bb7a2f8f3626815b7b8 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":1},{"id":"12db2fa46bd57983f09399d28c9f9dea176968ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 19:47:05.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9e39d33c40046848cfb1bf554031b6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528edbba912a4d09b7ed7d29597c2640 : FAILURE in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9809661e1dc4100a221ebbffca678b1 : SUCCESS in 11m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/846d990ebd254d8aa5488e8915ad7be2 : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73594879afc94e159c83c60ed28ef5a1 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5b01b3a30c466baeb0756eba166543 : SUCCESS in 14m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/650fb4237bab4c99a90644272fce3930 : SUCCESS in 41m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6db8e165e6464592b3232fe93d113e14 : SUCCESS in 1h 11m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/472cd1bf22974595b27fb06f4ffd8cf1 : SUCCESS in 1h 43m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c58e37c73764486a7b15bd0fd06d223 : SUCCESS in 1h 07m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d9f25274c70482caf28d2b2de616e1a : SUCCESS in 15m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b119f4deef29464c8c94a5828e0b98da : SUCCESS in 17m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7b6d47fe26354847ba41f162244b08dc : SUCCESS in 5m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3662d09e8d54adb9fadca6f5ce9287a : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4fc52504fd00486dba154e6400de33c6 : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/260b85e246a94b9285a2a676ba8c1823 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3fd3c1642a924c89b37dd9cd94caa955 : FAILURE in 46m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dc90de17ec5c49ba958f6743bfa0d7f8 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dcdfe5bd17f4435b89367313bedcb930 : SUCCESS in 1h 44m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/90bc5282135f4aa5adab9d5432babebf : FAILURE in 29m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e3cc339f1c94b67b54133e6bcf531ba : SUCCESS in 2h 09m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/01a38547d8f24afdb11dd3fa6bc06fd4 : SUCCESS in 1h 52m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/84d8517737ba4ee6aec295099878754c : SUCCESS in 20m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d3347b0f68b34c30956181f2a11dd292 : SUCCESS in 1h 19m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/92d31fa59e964e5fa760335a36e5e61b : SUCCESS in 1h 31m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a95d8394e54144308705d82fe2aace9c : SUCCESS in 38m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/96b270fe3c59498a92187dc20825baa0 : SUCCESS in 1h 20m 07s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/627bcc07d8484390a3a6c3fcfc37f857 : SUCCESS in 2h 10m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c7a42da0bca94a76af98b4d61e51f2dd : SUCCESS in 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c42d8c11d2f423b9e668b383dbde9e7 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"773f33b1ed9ad7fd1ab839e2e155fbbdd1693ebe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-04-13 21:47:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/984410 : FAILURE in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1f5b7626f7cf0b34fb23c13a8beb42d7509f7102","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-04-13 23:19:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/702505e3369c47f29e8758b9314f8e77 : SUCCESS in 46m 40s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/984410/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/984410/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"6b864872aca18c739fa8afac813dc7bb88bdc473","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 23:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09bc49ae0a65404cbcba1dd76d43e751\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78bc7ebccefb429388f342aa92545276 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b70e16cb01cf4964b705ab4e91696127 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df0382ab6a8846aba9f20a31de6ada1e : SUCCESS in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdadfee9501fc25f4615d3e54fd4086a65044daf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 04:11:24.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3607/.","accounts_in_message":[],"_revision_number":1},{"id":"64f266eb276dd86fef6d1c1f2ea34639e18f6b5d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-04-14 08:43:54.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3783/.","accounts_in_message":[],"_revision_number":1},{"id":"e207edfe3d19fa699d06c76aaa0d9048fa400224","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"date":"2026-04-15 11:19:13.000000000","message":"Abandoned\n\nDuplicate patch","accounts_in_message":[],"_revision_number":1},{"id":"f0f694a2f4ed94ce91f362bd38bd169750af4f62","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:55:30.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06a0ceb510673f9d34f70aea28cb92c23ef0e5f0","revisions":{"06a0ceb510673f9d34f70aea28cb92c23ef0e5f0":{"kind":"REWORK","_number":1,"created":"2026-04-13 15:14:17.000000000","uploader":{"_account_id":38319,"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","username":"rutuja.pawar"},"ref":"refs/changes/10/984410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/10/984410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/984410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/984410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/10/984410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/10/984410/1"}}},"commit":{"parents":[{"commit":"efa2e33d53c76497ee02b1d598c005fad893c746","subject":"Merge \"[Pure Storage] Fix performance stats collection error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/efa2e33d53c76497ee02b1d598c005fad893c746"}]}],"author":{"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","date":"2026-04-13 15:08:46.000000000","tz":330},"committer":{"name":"Rutuja Pawar","email":"rutuja.pawar@ibm.com","date":"2026-04-13 15:08:46.000000000","tz":330},"subject":"IBM-SVF: Fix force detach volume bug","message":"IBM-SVF: Fix force detach volume bug\n\nChange-Id: I7d7fa80af2f910c7a6b804d323309b42fa2988d8\nSigned-off-by: Rutuja Pawar \u003crutuja.pawar@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06a0ceb510673f9d34f70aea28cb92c23ef0e5f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/06a0ceb510673f9d34f70aea28cb92c23ef0e5f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
