)]}'
{"id":"openstack%2Fcinder~987177","triplet_id":"openstack%2Fcinder~master~I2df84091848978250741ed29f582222738b45f97","project":"openstack/cinder","branch":"master","topic":"bp/dell-powerstore-snapshot","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-04 14:46:31.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}},"37784":{"account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"last_update":"2026-05-06 13:05:50.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-31 07:34:34.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38286":{"account":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"last_update":"2026-08-31 07:54:58.000000000","reason":"Someone else replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-05-04 14:46:31.000000000","reason":"\u003cGERRIT_ACCOUNT_37784\u003e replied on the change","reason_account":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2df84091848978250741ed29f582222738b45f97","subject":"Dell PowerStore: Add manage/unmanage snapshot support","status":"NEW","created":"2026-05-04 14:43:16.000000000","updated":"2026-09-02 21:29:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28f734c986a87f823434a4c99174b26aba8d4315","_number":987177,"virtual_id_number":987177,"owner":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-31 09:55:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-09-02 10:31:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-04 14:43:46.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-04 14:46:31.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-05-04 14:46:31.000000000","updated_by":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-05-04 15:00:03.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-04 15:02:20.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-04 15:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-04 15:30:21.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-05-04 15:46:11.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-04 15:48:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-05 01:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 01:21:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-05 03:00:48.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-15 09:10:08.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-05-15 18:24:05.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-19 19:56:03.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-08-31 08:07:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-31 09:14:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-31 09:19:08.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-31 09:20:07.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-31 09:20:28.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-02 10:31:47.000000000","updated_by":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"reviewer":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"state":"REVIEWER"}],"messages":[{"id":"c6e19f70ed1eb9e3e430dd3086eb968ce55f7f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-04 14:43:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d59f90d9044f415f81af77f9f293c074e326444a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-04 14:43:46.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 987177,1","accounts_in_message":[],"_revision_number":1},{"id":"958a5915f9614f9e7e799ebf2c0f42c79ca77353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-04 14:44:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"40a35a532b4d5099422d38f729c1b8b14faa15f6","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-04 14:49:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abdce04c7af25ae80e9905f7114ba961fbce58a1","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-04 15:00:03.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/987177 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4fcd5cba3232b241babc006c362d5faa5ea1c6b5","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-04 15:02:20.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/987177 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"150ab8011e2b48ff317da734f048c5d8cff38f0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:06:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d302217e2fa4468da1474fe6b2a324b3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01d92bcb810b420193310147cfe25367 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68103e829bb7455fb3d9cabbcfd9d3d7 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2367522f984b4958b720551a81da2dce : SUCCESS in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e733bf87253d57949845f586152ac71bfeafff82","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-04 15:30:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/d73c69aa0db9448dbe652dc1ad8e101c\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/babeab61d7b24f65b897484dbab18b21 : FAILURE in 28m 51s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2284cdedfcb040a6b0d4375e371f003a : POST_FAILURE in 39m 48s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/6542c758ebde4b2d893150bd8722a98a : FAILURE in 40m 47s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bdcfde03985947a29af595517cdb623f : FAILURE in 8m 11s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d35de90ead89405d0129c7ff1745d5c3708e036a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-04 15:41:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/f15b637a82af4574a76725968a8b017d : FAILURE in 56m 55s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/987177/etc/tempest_log.log","accounts_in_message":[],"_revision_number":2},{"id":"5ff4b5f2729a44b824801325420ab2a408d645b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 15:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4062/.","accounts_in_message":[],"_revision_number":1},{"id":"1fc412da9cabc9dc63821cb2e0aba17ff7ec1af7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-04 15:48:28.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2df84091848978250741ed29f582222738b45f97 .","accounts_in_message":[],"_revision_number":2},{"id":"227cec89d594f619224265182d7cf1a28177dd07","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-04 15:49:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2df84091848978250741ed29f582222738b45f97 .","accounts_in_message":[],"_revision_number":1},{"id":"0d4dd6a7f3de628436b35940d8b1005f67720c61","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 15:55:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3886/.","accounts_in_message":[],"_revision_number":1},{"id":"9eee6cf892e30c135ba5c241a3305bbb9358a161","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 17:05:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3887/.","accounts_in_message":[],"_revision_number":2},{"id":"e9144b85517a4490db5d7884a0ba3b65611fbbee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-04 17:07:12.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4063/.","accounts_in_message":[],"_revision_number":2},{"id":"7c39dad14eac21fece4fdc145009221786f89cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 01:21:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73216dcbd1724ef4979ceebf62dcb311\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e63d097378c34a898cdcf559449a2c48 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8d23270295445a8ba8e484f5e7de77a : SUCCESS in 9m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/071ee546232349c6891c85365ceef56e : SUCCESS in 6m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bc5de39adab46f9bc1e2d82d39687ac : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e7488de40a47eeb0843e06fdb3077f : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/b9d28f38d343456c8169393254d1cc70 : SUCCESS in 39m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d5867afc786649c1ab4d3ccca79ebd49 : SUCCESS in 34m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e830d60b01a4c929eb3ab7420806c1b : SUCCESS in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47fe637734d742208e40d4208fe5b54c : SUCCESS in 1h 07m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8950d25bc28c4de0a81a82bfdb0647e2 : SUCCESS in 16m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d3309b384274030be08f4dc4aefce58 : TIMED_OUT in 40m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9a7fd4bf268f495ba1cd20b950fc94ef : SUCCESS in 6m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5b75ebb89284589975508d5e7973889 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b9f494415f849e884012876b79b86ee : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/87d11c98c2e844c5b97d74875bdefeaa : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/95644d8fcbaf4dc698e82842cc192faf : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5fafdf5c75954a75a9b9c2f3f9fd7db3 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0a896c079cc446f91e9e46d74fbf976 : SUCCESS in 1h 50m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2c19088507ed4fc593929d181afb143e : FAILURE in 2h 47m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5b306ac140648b38addad57bcfd9c31 : SUCCESS in 2h 06m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e19789d2b8d3449887e541fe0d9378e5 : SUCCESS in 2h 01m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3d5b4cfef9124ce19bc3c8ef3ecc7b04 : SUCCESS in 26m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/730834ab901740f480222a2b2f0f903d : SUCCESS in 57m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d28f1e2e7ac1481ba1a792e76edb74ce : SUCCESS in 1h 33m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39ef1ad3747e4444887d2f90f6dd8071 : SUCCESS in 44m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf83cd7ee67044588937df89c4b6c38b : SUCCESS in 1h 41m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/52290a25140e41468cf3a34ae3703230 : SUCCESS in 2h 15m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d82ee63f47aa4ef4a31c5d6a4ab2907d : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6e48db13b99f9ebee4365c79a52a7e8c7961b09c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-05 01:30:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_05_2026_01_24_36\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"0e0918aa764fd53196cde7b8473ca14cfebc0295","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-05 03:00:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-77-987177-2-fc : SUCCESS in 1h 22s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-77-987177-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"729f5c15f834eedf314eccc73cbdf76ada863c6a","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-05 06:55:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff25e7684b9d2c163224d3baff11c8d0b3736cc4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-05 13:59:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/17963e41ecb14200a5f45369ed62de8e\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/03a2aee7638f4f23b7c031f2b9817a7b : FAILURE in 3h 31m 28s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9117ae7afcd044bb92cd9825472c7a0e : SUCCESS in 2h 24m 49s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8d2f34ab3ba648fd8eef578a81acff46 : SUCCESS in 2h 11m 48s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/751931d144ec453999140ea06c0c1a12 : FAILURE in 7h 02m 52s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"455c4e0724d3bb8b094131518bde2325eef925f6","author":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"date":"2026-05-06 07:21:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0a7d1de3a2eee2b71c765f0fd67a6e24de0c28f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-05-06 13:05:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/68dd57b2586f46ec83092aabe623e472\n\n- powerstore-nfs-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/d0859676aaac48c1aa548bfcf734c469 : FAILURE in 2h 34m 16s\n- powerstore-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8c1ffdb4be86407fa4f446deb398a89e : SUCCESS in 1h 23m 22s\n- powerstore-iscsi-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2b4cf91216c64550a12a693070df0779 : SUCCESS in 1h 27m 45s\n- powerstore-fc-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/61dad1d1d5434dc8ae6a1cc15d98e31a : FAILURE in 5h 43m 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"3401fcad4a787c92b554157b6d7585c2194012cb","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-15 09:10:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe6b5af36c15e9ccce068bcf05dfa9ccd76855fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-15 18:24:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/214750dacf50410f815fb213efc3fcec : FAILURE in 28m 58s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/987177/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/987177/etc/tempest_log.log","accounts_in_message":[],"_revision_number":2},{"id":"dad8a24f66d7deae1dbad46512e97c89ac0075ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:56:03.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"66f849923072c9594ee7122f4ac1d8fae9fd6ed9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:58:10.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"caf692f1a03cbc6a4c0d2dd71fc860aad9941547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-31 07:34:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3ddc02aaff5a951fe47bed630a83882ac45a5802","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-31 07:54:58.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5330/.","accounts_in_message":[],"_revision_number":3},{"id":"332df3976e3a954f0194594434f1c2ed26b93c4d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 08:07:53.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/35e997c5de5540b3aaa72865d3d093a6/","accounts_in_message":[],"_revision_number":3},{"id":"53f4c13c46843e094abfcb9ade62c62d3684eabb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 08:17:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d04b82623dd4de78a6d6a49f7b2900a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff57f125294d40e7949140751b1a5346 : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84fab00ce1294f5ca552c9cc624c502a : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/857d32734bcb47b4878c983814dc1828 : SUCCESS in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8cdef1b52ef1e2401f7924124f8830b85d8432cb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-31 08:27:39.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2df84091848978250741ed29f582222738b45f97 .","accounts_in_message":[],"_revision_number":3},{"id":"a09f782ed58bb3d972be9ff4f743026e8542c506","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-31 08:54:14.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/35e997c5de5540b3aaa72865d3d093a6\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fb12ba6e368848918c862c6c099d4dc0 : FAILURE in 45m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a54e2d1ce92d4abd8f8993fbe9b2ee08 : SUCCESS in 33m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0fe1f8a27840479999ca1066f405a597 : SUCCESS in 46m 14s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fbfb9b26f2854c1682e93a4e973a51f5 : SUCCESS in 35m 18s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c23cdecf1d4c47da9f6be38913ab80ca : FAILURE in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f378c4a6a36ca8dc6f437db391f10c024e74c9b3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-31 09:14:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9e2dab7b27b341d687eb56d6482303dc : SUCCESS in 46m 05s\n- cinder-storpool-tempest-iscsi-multipath build/69ec7f81869542f5b3a1c55962dcb757 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c28c4ff13ceedc11cffa8ed5a2c8183344960620","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-31 09:19:08.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1375 : FAILURE in 0h 1m 51s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/987177","accounts_in_message":[],"_revision_number":3},{"id":"43bfd823ce399e67e8f2aa11bc39098e35e259b5","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-31 09:20:07.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/1106 : FAILURE in 0h 2m 7s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/987177","accounts_in_message":[],"_revision_number":3},{"id":"94b89805f63615527f092aa1a6a9e1f17ce2316b","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-31 09:20:28.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/1014 : FAILURE in 0h 2m 16s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/987177","accounts_in_message":[],"_revision_number":3},{"id":"f249ac9b84ee757e2d24c63b41407eab4187d3c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:55:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cb796eca484a7c85da4e885fcb9055\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17fd4a61bcc24df1b723c61fd0924b87 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b466098aeb7f4ca0b0bd6c3649416e03 : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26de113b6a994e83bc5d8ba354e14d84 : SUCCESS in 11m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25127d0383714997bd7bf3d028c128f6 : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81d4344d4d94fa593aae6ffebd7cf5d : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3631175c67834a5b92bace493407b880 : SUCCESS in 1h 12m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a5417536d0c4d7aad95b8fa3365b183 : SUCCESS in 1h 03m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f51f5c16f39547109a91df91e7f73a9a : SUCCESS in 1h 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fa93999200546e8b3f1ead05dba43dc : SUCCESS in 48m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d1e49121936445d896a1bf472a149a8 : SUCCESS in 7m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f177d2a1c170426aa2e9892ff148e813 : SUCCESS in 20m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/83d5bc58b9ee4fa89118846569cd809e : SUCCESS in 6m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b96cf261c7074037aaf1cba9ef02b159 : SUCCESS in 3m 49s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bebb5bd0cd8d46e5b8480d3c148f7492 : SUCCESS in 10m 35s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d221160dda0b4d059071ecb96b81ea2e : FAILURE in 7m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5db322b465a4a80a9c0f7016da4e48d : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/63ec005b5405441fb4c048a566c85d51 : SUCCESS in 8m 03s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0e7a27dd87514d1f930bcffbbc87f88a : FAILURE in 4m 03s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ea013a1a619e45ef97d458a352d33358 : SUCCESS in 6m 52s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a1abc3981ee84e4ca2e2ab50de5e234b : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3526327d281b4a2f9714daec3b789cb5 : FAILURE in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/377aebba1e3c49879be0a4dfec1ec409 : SUCCESS in 1h 26m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4eb7b4e03a93449eaaeb7a35d9be9784 : FAILURE in 41m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2722be68accd4745ac2a5fb75cf85829 : SUCCESS in 1h 51m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6cae0c56607440cbaa8f41173acdd3ae : FAILURE in 2h 15m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/45146f3246ad48578169d263db7872e1 : SUCCESS in 1h 39m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2ba06acc83534af5a34c35c70b376876 : SUCCESS in 39m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d141e9a1609b4882ab897f05ba2ffcd8 : SUCCESS in 51m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ba22db061924e518bc7e7e660f3ceb0 : SUCCESS in 1h 18m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2973b038512a4b09af4e318639c24a4d : SUCCESS in 25m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0eda4b55186447a812786705a380932 : SUCCESS in 1h 41m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ad3a4add0024f839679b03131c9428f : SUCCESS in 51m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"ebdb9698dd240cdb144ee7826a8788d0fc4b9cb1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-31 10:11:28.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_08_2026_10_04_04\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"d65a573331e3cfac5bcf61c106a0e7e3b3186979","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-31 16:44:28.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-77-987177-3-fc : FAILURE in 14m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-77-987177-3-iscsi : SUCCESS in 1h 18m 55s","accounts_in_message":[],"_revision_number":3},{"id":"fbfee2957452803356fa2720251d586b1db6508e","author":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"},"date":"2026-09-02 10:31:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28f734c986a87f823434a4c99174b26aba8d4315","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 21:29:10.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5154/.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"913b2669b5b80bf6d8512279e04db06f8d14c69f","revisions":{"a67eb14123702f4d52b3e7adeffa9ba9476b1c42":{"kind":"REWORK","_number":1,"created":"2026-05-04 14:43:16.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/77/987177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/987177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/987177/1"}}},"commit":{"parents":[{"commit":"4015ff29723f24363af4d6aa6d0d0f7bf5c4ec4f","subject":"PowerStore Cinder: Added support for manage/unmanage volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4015ff29723f24363af4d6aa6d0d0f7bf5c4ec4f"}]}],"author":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-04-09 20:35:27.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-05-04 14:28:54.000000000","tz":0},"subject":"Dell PowerStore: Add manage/unmanage snapshot support","message":"Dell PowerStore: Add manage/unmanage snapshot support\n\nAdded support for managing (importing) and unmanaging existing\nPowerStore snapshots in Cinder. This includes:\n- manage_existing_snapshot and manage_existing_snapshot_get_size\n  in the adapter and driver\n- get_snapshot_details_by_id and get_snapshot_details_by_name\n  client methods\n- Unit tests for all new functionality\n- Driver documentation and release notes\n\nImplements blueprint dell-powerstore-manage-unmanage-volume-snapshot\nChange-Id: I2df84091848978250741ed29f582222738b45f97\nSigned-off-by: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a67eb14123702f4d52b3e7adeffa9ba9476b1c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a67eb14123702f4d52b3e7adeffa9ba9476b1c42"}]},"branch":"refs/heads/master"},"c9a127832e2bdce80bf25e11cce8433f760c6869":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-04 14:44:25.000000000","uploader":{"_account_id":37784,"name":"Abhishek Gupta","display_name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","username":"abhishek-gupta14"},"ref":"refs/changes/77/987177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/987177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/987177/2"}}},"commit":{"parents":[{"commit":"da3716f3583a406a81e3a7ff1aae20e4fe2bdcc2","subject":"Dell PowerStore: Added support for manage/unmanage volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da3716f3583a406a81e3a7ff1aae20e4fe2bdcc2"}]}],"author":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-04-09 20:35:27.000000000","tz":0},"committer":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-05-04 14:44:25.000000000","tz":0},"subject":"Dell PowerStore: Add manage/unmanage snapshot support","message":"Dell PowerStore: Add manage/unmanage snapshot support\n\nAdded support for managing (importing) and unmanaging existing\nPowerStore snapshots in Cinder. This includes:\n- manage_existing_snapshot and manage_existing_snapshot_get_size\n  in the adapter and driver\n- get_snapshot_details_by_id and get_snapshot_details_by_name\n  client methods\n- Unit tests for all new functionality\n- Driver documentation and release notes\n\nImplements blueprint dell-powerstore-manage-unmanage-volume-snapshot\nChange-Id: I2df84091848978250741ed29f582222738b45f97\nSigned-off-by: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9a127832e2bdce80bf25e11cce8433f760c6869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9a127832e2bdce80bf25e11cce8433f760c6869"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a67eb14123702f4d52b3e7adeffa9ba9476b1c42","theirs":"da3716f3583a406a81e3a7ff1aae20e4fe2bdcc2","contains_conflicts":false}},"913b2669b5b80bf6d8512279e04db06f8d14c69f":{"kind":"REWORK","_number":3,"created":"2026-08-31 07:34:34.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/77/987177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/77/987177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/77/987177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/77/987177/3"}}},"commit":{"parents":[{"commit":"11d0fbfcf7364e2e6a0c95bc1b687697b7907787","subject":"Dell PowerStore: Added support for manage/unmanage volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11d0fbfcf7364e2e6a0c95bc1b687697b7907787"}]}],"author":{"name":"Abhishek Gupta","email":"Abhishek_Gupta14@Dell.com","date":"2026-04-09 20:35:27.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-08-31 07:25:30.000000000","tz":-240},"subject":"Dell PowerStore: Add manage/unmanage snapshot support","message":"Dell PowerStore: Add manage/unmanage snapshot support\n\nAdded support for managing (importing) and unmanaging existing\nPowerStore snapshots in Cinder. This includes:\n- manage_existing_snapshot and manage_existing_snapshot_get_size\n  in the adapter and driver\n- get_snapshot_details_by_id and get_snapshot_details_by_name\n  client methods\n- Unit tests for all new functionality\n- Driver documentation and release notes\n\nImplements blueprint dell-powerstore-manage-unmanage-volume-snapshot\nDepends-On: Ia2b136acc8229d27e16d4363be16de4041bb7dd8\nChange-Id: I2df84091848978250741ed29f582222738b45f97\nSigned-off-by: Abhishek Gupta \u003cAbhishek_Gupta14@Dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/913b2669b5b80bf6d8512279e04db06f8d14c69f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/913b2669b5b80bf6d8512279e04db06f8d14c69f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38284,"name":"Prasant Padhi","display_name":"Prasant Padhi","email":"prasant.padhi@dell.com","username":"prasant.padhi@dell.com"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
