)]}'
{"id":"openstack%2Fcinder~988246","triplet_id":"openstack%2Fcinder~master~I6934c4332b442b32fd6a50bd1c0152cf02dc469c","project":"openstack/cinder","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-15 19:13:50.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-06-10 12:29:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I6934c4332b442b32fd6a50bd1c0152cf02dc469c","subject":"objects: Remove aliases of objects","status":"MERGED","created":"2026-05-12 11:47:11.000000000","updated":"2026-07-07 09:12:52.000000000","submitted":"2026-06-15 19:13:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988246-typing","meta_rev_id":"b384e7ccfc40adaf540f19d54e3e7550fc723664","_number":988246,"virtual_id_number":988246,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-15 19:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-06-04 18:05:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2026-06-10 12:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-06-10 12:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 12:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 12:19:30.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-12 12:27:52.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-12 12:41:25.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-12 12:43:52.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 18:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 18:01:09.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-12 20:14:57.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-16 17:19:41.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-04 15:18:11.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-06-04 16:25:44.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-04 17:14:37.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-06-04 18:05:56.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-06-04 18:07:06.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-09 22:42:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-06-12 17:57:42.000000000","updated_by":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"reviewer":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"state":"CC"},{"updated":"2026-06-15 19:13:59.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"6c0eb8b5ea8b3b9980ef9b36ea809c87df0b072b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-12 11:47:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4128a967bb580e8312d5ac3d80ea94bc5369fd1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9970a9e038fc42ecbaa5ad5272a34572\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4280a63c1c7a462b884088ffc994075c : FAILURE in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc600e4d5d9a4617b9fa4367b23502ca : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cdfcc3e1fb894c1abc97fc42e4cd7c19 : SUCCESS in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16de713ca004a9e3e85b357dc325d4bd915c46c1","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 12:19:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/f39cc24c3b214f9cbee073dfc492d57d : FAILURE in 26m 14s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988246/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"faf354c44b3c171b1cc43cd5ac3a42d33e3db4a6","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-12 12:27:52.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/988246 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"57539b2c13ffa108e14fbad097a8fbaa13339289","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-12 12:41:25.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988246 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a8dec6fd2336f14141507e6cc671413631fe25e7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 12:43:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6934c4332b442b32fd6a50bd1c0152cf02dc469c .","accounts_in_message":[],"_revision_number":1},{"id":"bb349bd87d125552348a07500825c4c9f977d9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:00:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd64581d3d74745923bd060ac4a59f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5db2370b045401fbe557a6dd7919bc9 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c237a8cb99b4ede89854ba2a7cdb74c : SUCCESS in 9m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7951bf1ad874947a77be22b98340554 : SUCCESS in 7m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d06f576eba114314845c309fa9e73ea6 : TIMED_OUT in 42m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc43f406e844798ae9895d6d6317b5f : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/6fc6028f5c5040fb94b38e7615c37f0e : SUCCESS in 1h 01m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47399afba73142398203a4b2b5e8547e : SUCCESS in 1h 11m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ed1f37869844bb8b5074fdbab932a32 : SUCCESS in 1h 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7eaa9e643ab48498964fda74dba5fab : SUCCESS in 59m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b29b2b6a18f343f68ab394eceb88b18b : TIMED_OUT in 41m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ca5d1bb0dc2435688769c427d17ff7a : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/540ade99df824435bcc8543afce9d234 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/75ea289abf70452ea4fad0c78759257b : SUCCESS in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8b2787668f1d4f67a5082f7fc3647a7b : SUCCESS in 6m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6a71f3e27ca84c469e05568c952f5135 : SUCCESS in 9m 10s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/56bcee5b2b1d42cb946b88376eb4eaa5 : SUCCESS in 40m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d12f331aa8e64b509e9187a059675eff : FAILURE in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac198ad624a94d479a7d789ab41f751d : SUCCESS in 1h 46m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/55d56085eef445b88d256f0fd17bf069 : FAILURE in 2h 24m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/575ce25e354646c481a1730e8151fad0 : SUCCESS in 2h 34m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c3b174940e2942d69a1cfc923e64bbd7 : SUCCESS in 2h 02m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ebabc67f40454810bfdbf76e9c3f677e : SUCCESS in 34m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/17490f6c3e3b473e843ed5d501501267 : SUCCESS in 1h 12m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5713969bfc284773a86c3171ec03d338 : SUCCESS in 1h 51m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3e128f44089e46489f55e3ce1d493d18 : SUCCESS in 45m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e99f375c26e346709a0976cacfe4aa1b : SUCCESS in 1h 44m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc47044dfda14c458ea18c8cc8feb240 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bcc4c06dea2db2acd5f943b8f79e01206898fcb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-12 18:10:37.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_05_2026_18_04_09\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"2eb48b876aa91c287a5386231601a00f7d85ecb3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-12 20:14:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988246-1-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988246-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"0320d53a7f55cc7822aae57065ad9cc84af383ee","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 17:19:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/899641354cd8410e899b092163eca597 : FAILURE in 29m 50s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988246/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988246/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"627824538c9b7305b14598ed8ad2706b4203f5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-04 15:17:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"39182f0135f1484705ba837f4b36f28129dfa6bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-04 15:18:11.000000000","message":"Patch Set 2:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/838/","accounts_in_message":[],"_revision_number":2},{"id":"15a2e2c871cbc73d6943b3f6330c4d44f343831f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-06-04 15:18:15.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/988246/2/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/988246/2/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":2},{"id":"6ba8446ff3d8c4b29db8802c83fdd37ad211225c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 15:53:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2093aa54b43b4f779e7a101caee7c031\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/631ac6500ce94f4e9add52b3b7534181 : SUCCESS in 17m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/655523f4b9914e09803938b48dac26f7 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2695c1f3324642bea298e7a6fd7cd68c : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff1e13ac457f5c99b31852093fc97531887dc15f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-04 16:13:11.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I6934c4332b442b32fd6a50bd1c0152cf02dc469c .","accounts_in_message":[],"_revision_number":2},{"id":"97f4c525a9c59dac86d9a371ba736a7f6f8f00ec","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-04 16:15:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/353f7edc09f34fad8d5d591a647ab921 : FAILURE in 30m 37s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988246","accounts_in_message":[],"_revision_number":2},{"id":"3eb7efd0f33a3e7e7c05d613cfecaa2bbfb6818c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-04 16:22:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c9e0853d2ec1470dbe6f4bce7c8be8e0 : FAILURE in 37m 08s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988246\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988246","accounts_in_message":[],"_revision_number":2},{"id":"4a7a2ebbf0c5916dfda40283da13f25b7ec61512","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-04 16:25:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/c2d73b89df0240fc9529b3444f6e5565 : FAILURE in 36m 59s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/988246","accounts_in_message":[],"_revision_number":2},{"id":"7030e17fb01206e2037420c68b4baaf02a94e0f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-04 17:14:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4360/.","accounts_in_message":[],"_revision_number":2},{"id":"de868389d8cf18732eb6915549397f59720892cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:57:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552567c5be574863b7829cef5c350d00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d92558876f4959a04fb3b7e92dc321 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aa9c9c2361d4ff4bca4ebe59dcce482 : SUCCESS in 9m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75ccb767a116406482a33859896ba721 : SUCCESS in 9m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2063c8a9b79b4cadb1287e83076d8a39 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc8d73120124ccf83f2b133b7c3efd2 : SUCCESS in 15m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/36717b3f3b1248368af48c6c1531ee16 : SUCCESS in 1h 08m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1cb3642b3b214a279ce5de6a7850f9ab : SUCCESS in 1h 07m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c3021767bfa4b24b1c58c84c66ddd34 : SUCCESS in 1h 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dd6a0d5d80f4b9bb530d52256b94ae4 : SUCCESS in 36m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60d84e6141546718bc8db29e0abd4be : SUCCESS in 9m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0da6fb57a47b4de7902c18340c5ab114 : TIMED_OUT in 41m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d4cabd8371e94357b17827da878b3a18 : SUCCESS in 6m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6bfa704e46394712b46ca7755563a948 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/921740a999894043888c86ecb734383b : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/51d1be9e610346ec8b34f84363758cd2 : SUCCESS in 6m 55s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/27b1d17081ba49b191439483d33b0f0e : SUCCESS in 11m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecfb7c0b412446ca9011b1eb155f5453 : SUCCESS in 31m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c0a900555a64bfab756e858c91c6613 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/94a9a637de1a489f8d3bf5d3e66346a8 : SUCCESS in 1h 51m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6a258a1cffa8437bbfcd9640f12e9192 : FAILURE in 1h 40m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/801c09f178014f65b717d639d85c742a : SUCCESS in 2h 32m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/064067c71b3e4270a192ac515e4b40f5 : SUCCESS in 2h 01m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/201b38a8d0b64daf89ac57a67cf95cd6 : SUCCESS in 23m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55739c50ed9e47aeb06381de901d9efe : SUCCESS in 1h 04m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ce6dc9a1b71a4c4daae35d60a511bb1d : SUCCESS in 1h 35m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7034a940dcdf47bb94096a15f930c602 : SUCCESS in 48m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88c6c8bd68994dfa9b38a16f567679c5 : SUCCESS in 1h 34m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97b2dceca641476d89da0ee1cd8c8864 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c008bac86a0aa629613d240836f2b5c585ad49ee","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-06-04 18:05:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57789b826e49e0913cfb9218f49e07f28f7fd3eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-04 18:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9435f3c123134db7b3a7c1cd37424d41 : SUCCESS in 1h 00m 54s\n- cinder-storpool-tempest-iscsi-multipath build/b3dbfa55d99d419f85caa41fce499129 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":2},{"id":"bde3ed70855080cbe1260d5cb76bc223597d2bf2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-04 18:08:02.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_06_2026_18_00_32\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"2e91141d64e666ae8702f0186cd82b6fbd6eb4e6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-10 12:29:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b8b1b1cc153d5d870f37830575c7ce781efdbcd","author":{"_account_id":37025,"name":"VASTData CI","display_name":"VASTData CI","email":"manila-ci@vastdata.com","username":"vastdata-manila-ci"},"date":"2026-06-12 17:57:42.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/985039 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"9222fdf4b3da1c55571b6bb588a6a605ecc439ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 17:58:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77fc71535a79b6ab386955ab55e7095d6103819b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 19:52:15.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/cinder/+/985039 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"9e45b71d9ea47173725a03638ee88d9b8372d942","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-15 13:16:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5491ce2b15a946896ca171a9818a4ae8d130f59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 15:23:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0692f5ced6b4b32997d824b75ea60ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3d3a3f698f4db882ab5a0aa4518649 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eeb53f3e6cf34c24956119471b2a6174 : SUCCESS in 10m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/776f70820ce64b8194c4ee7aaeff59e6 : SUCCESS in 8m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06ae5d24152847609203759fd74d4f44 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b49a8af4bd14abf96e36a75531fa907 : SUCCESS in 13m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/438fda47b9f24be0b9a1487c487ee114 : SUCCESS in 55m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f7e3befc1724d1197a97eaea2920871 : SUCCESS in 38m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9b6fa1d9fc343d0ba671246c9686485 : SUCCESS in 1h 49m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2123a8b505fa45b19fac9f37408deab5 : SUCCESS in 1h 01m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65ed2be8dc5e4a78acaaed783946636f : SUCCESS in 9m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4583a95244340178aa9974aef96a64c : FAILURE in 25m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac1bdb9e53e44c8aa25ee0f6493ed575 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/380fd05a23f141ad9f8772f812dce91d : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68ab6bafd4d14fff8efc6f1bd0361001 : SUCCESS in 8m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9d2f4b9b77e74517986e0fe88d5498d8 : SUCCESS in 5m 55s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9fb175c55df54ebcb272086779c68ee7 : SUCCESS in 10m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef2fcf2f4998421c921103b0604ccd10 : SUCCESS in 29m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e751f4548af4b3ba44a8fa51a308b8d : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9e88fcfd64b94cd8a38dda131a072d1c : SUCCESS in 1h 08m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4c189a0f04c54a079a5fcceb70fa46be : FAILURE in 1h 38m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65a22956911b4ca386ccea08bb82b6b1 : SUCCESS in 2h 00m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/af8fceb37fb24473a01d6ccdf2339b85 : SUCCESS in 1h 46m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/975289280fb34d23bbaaf40a4cc1a148 : SUCCESS in 36m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8ff72f3a03094d88b1672b5ced378330 : SUCCESS in 39m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/db5e0b9a6dc2413697965bbc5aa19f51 : SUCCESS in 1h 39m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f11bce1676345e1869993f7910448de : SUCCESS in 40m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9599adb3a0164015834aa06a0df8125d : SUCCESS in 1h 52m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b99e1280f2564160aecb342e7e030c58 : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"904e908eb37278b1c48257d0a75184f8bc8b4f36","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-15 15:34:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_06_2026_15_27_02\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"99ee2c3acf70bd16b5f6aadc7b58bc9b8706af0d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-15 15:44:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/52dfc4b8ede34d44b2ba615250096249 : FAILURE in 54m 59s\n- cinder-storpool-tempest-iscsi-multipath build/ccfae32b24f74789858c43da324de6a5 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d6b1a2f535d0c0afa5a72914f41335b487234186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 16:20:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4583cf6b85df524471cdab1722d0e200b34cef52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:13:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8ff038455e4d2194f27cc7777f2072\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5768626c598a4dd7a5894d9583ec52ba : SUCCESS in 8m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3de266fb1cf4e389a328274fb2f2351 : SUCCESS in 11m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3dd494940b449fba2b728b0505428e7 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e89cf0ca084f9d8239993a5ab90177 : SUCCESS in 15m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/9109ad34b1ae4e619a59292458a13a78 : SUCCESS in 1h 06m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d23efbd9726740538b8ad475f3d7f3ce : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/011409c6baca406db34c2c4c693780c2 : SUCCESS in 1h 42m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8ecd887fabf4b98919eedffa73f878f : SUCCESS in 1h 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089a4003f3b04ccf92815143fc930c69 : SUCCESS in 9m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/720b9e4b599548138b1512b17612ee46 : SUCCESS in 1h 08m 13s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4213ddd0d15843f9bdf03039f1bd692e : SUCCESS in 1h 54m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71096f260bc747f4a0ccc2b324e3e891 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8f88993dfe1ea49ee6fd08bc599a5f848d9b0474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:13:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"02083441ed4661ff03b684eb8570a0af7cae813c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-15 19:13:59.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/2235/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0f3168bdaca5bf65e47803ac2a282fa2e61129fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:18:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ef46ab5a9249db91db1d8a55e237a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359d89b89eb647b189b32b8ab18ca0c2 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2},{"id":"eca3bed78155c95abcdf5dd7140e3b75badc11d1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-16 00:17:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988246-2-fc : FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988246-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a571328e21d28f6a4b92b5532f434a3aa6973174","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 05:57:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70b5a00d63f2486a8e50ae4bb3b92bc7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1eb7c57fd5a48dfb3361c13dbbbcf4c : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07fa76b3a4d642538733335001e33b6a : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b36cbf47a47244539d93580ea6ed5413 : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c508b4ac507901a68c3718707d7d7e5934e1b49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:00:30.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"b384e7ccfc40adaf540f19d54e3e7550fc723664","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-07 09:12:52.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4184/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2bce2c7698bbca225e897aa7381c298fc2922e2e","revisions":{"20324183911b651a957cac19d889228566abad12":{"kind":"REWORK","_number":1,"created":"2026-05-12 11:47:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/988246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988246/1"}}},"commit":{"parents":[{"commit":"c80c236c3c46629e7ce190c90f408722bd79f2ff","subject":"typing: \"Import\" objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c80c236c3c46629e7ce190c90f408722bd79f2ff"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-12 11:25:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-12 11:33:38.000000000","tz":60},"subject":"objects: Remove aliases of objects","message":"objects: Remove aliases of objects\n\nThese are unnecessary and confusing and type checkers don\u0027t like them\nbecause they see variables rather than types. Fortunately the things we\nwere aliasing are not used outside of cinder.objects.fields and a test,\nso we can just replace those references with references to the original\nfields from o.vo instead.\n\nChange-Id: I6934c4332b442b32fd6a50bd1c0152cf02dc469c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20324183911b651a957cac19d889228566abad12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20324183911b651a957cac19d889228566abad12"}]},"branch":"refs/heads/master"},"2bce2c7698bbca225e897aa7381c298fc2922e2e":{"kind":"REWORK","_number":2,"created":"2026-06-04 15:17:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/988246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988246/2"}}},"commit":{"parents":[{"commit":"8f3a53ad02cb077c687fa6e6479df31ffbae8807","subject":"mypy: Migrate to module-based configuration (1/2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f3a53ad02cb077c687fa6e6479df31ffbae8807"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-12 11:25:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-04 15:16:32.000000000","tz":60},"subject":"objects: Remove aliases of objects","message":"objects: Remove aliases of objects\n\nThese are unnecessary and confusing and type checkers don\u0027t like them\nbecause they see variables rather than types. Fortunately the things we\nwere aliasing are not used outside of cinder.objects.fields and a test,\nso we can just replace those references with references to the original\nfields from o.vo instead.\n\nChange-Id: I6934c4332b442b32fd6a50bd1c0152cf02dc469c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bce2c7698bbca225e897aa7381c298fc2922e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2bce2c7698bbca225e897aa7381c298fc2922e2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
