)]}'
{"id":"openstack%2Fcinder~988252","triplet_id":"openstack%2Fcinder~stable%2F2026.1~Ifd632433ce5759494172b74ce311816e10728d21","project":"openstack/cinder","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-07-16 19:23:52.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-28 14:55:28.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"Ifd632433ce5759494172b74ce311816e10728d21","subject":"Fix properties\u003d{} being sent to Glance","status":"MERGED","created":"2026-05-12 12:49:49.000000000","updated":"2026-07-17 12:09:51.000000000","submitted":"2026-07-16 21:24:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988252","cherry_pick_of_change":980692,"cherry_pick_of_patch_set":7,"meta_rev_id":"4d8200dac208c309eaae18c77defb65e5e9f4b62","_number":988252,"virtual_id_number":988252,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-16 21:24:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-07-16 19:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-05-22 15:26:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":2,"date":"2026-05-28 14:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2026-05-14 08:41:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-07-16 19:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31903,"name":"Alejandro Santoyo Gonzalez","email":"alejandro.santoyo@canonical.com","username":"alejandrosg"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 12:49:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":31903,"name":"Alejandro Santoyo Gonzalez","email":"alejandro.santoyo@canonical.com","username":"alejandrosg"},"state":"CC"},{"updated":"2026-05-12 13:15:22.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-12 13:18:36.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-12 13:25:48.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-12 13:47:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 15:34:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 05:29:22.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2026-05-14 08:21:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-14 08:41:11.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-05-14 08:41:11.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-05-16 17:53:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-22 15:26:41.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-06-19 20:01:20.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-07-17 12:09:51.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"c88d1233388bc230c770d7e3f56c2b5eb587bc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-12 12:49:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cfd6b64bcaaca7354a27471fd89b7f5f06ec6dda","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-12 13:15:22.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/988252 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"52d470e824bba5adf247a59d77953becd97854b8","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 13:18:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/230fa960486541afbbcc7541959feb32 : FAILURE in 28m 03s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988252/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"e8fb83e49a2737318d395874c6fd391cd137d8de","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-12 13:25:48.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988252 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"aa10149a8c6b8b8fdeb516b29c317c198c313866","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 13:47:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifd632433ce5759494172b74ce311816e10728d21 .","accounts_in_message":[],"_revision_number":1},{"id":"44d334ca30a63c3b4e1a40e86d775409bd4bd97c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:34:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88dafa4e10664deb82da70b855b3cdce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba116ed75ee240f58a82ba7e73791f25 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05d87b1cbd3248ce92593a2f3c5195fa : SUCCESS in 13m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5904203d50c9432cb14520d89c563d8b : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7238b1bbd2724839bd22f919c8dfb523 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/65d579a5261743b89f26c765ef913574 : SUCCESS in 1h 08m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de7f8969fd4d4e3a9a094d3d1009bc59 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ecc7a2228f0422b9322e5f3ff020972 : SUCCESS in 1h 26m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f8d04ffe30c44e398b16c134b2a191b : SUCCESS in 34m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d2eec40c48040b6b97aa5384421d1ce : SUCCESS in 16m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54550741ebf345cdb08247a615f149fb : SUCCESS in 22m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2829c6e52674f8d83e184b59ac61004 : SUCCESS in 6m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f32efc8368540dea22da0f694df3a16 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c001225ceadb4d7e93cdb139ee114c89 : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9e5f3a16878d4906a90c1441fccde202 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bde1d675148c47029ae484997c6f8691 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cdcde1cfd1e245bf9b189abda549f335 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f616aaa758f0474b98dc7fda6970ec68 : FAILURE in 14m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dfe850fcba0a4b08b082fe424f5fdf5f : FAILURE in 2h 38m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/570dbc7667a44898bc04d5680c251822 : SUCCESS in 1h 49m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0b3afc7fc39741b09176c7c62e96051d : FAILURE in 17m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4ead5f21d933457996249cc4ee78104d : SUCCESS in 23m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86fa524f27744b6e86cb96858e92c56b : SUCCESS in 1h 04m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/601d7818e6c2480085f1c52a7367ee57 : SUCCESS in 1h 28m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5377fac5df8d4dc9a81637e8aa9ff3fe : SUCCESS in 45m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/374508ef7e994f778c965251e4106e8a : SUCCESS in 2h 02m 52s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/6d6f48d86047469fa4f996c8fe2fc115 : FAILURE in 21m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59293c3eeadc4763bebc96795051475c : SUCCESS in 1h 47m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dafb31ea2c14e1db6585686f1ee3e88 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"be60d1109d740f95ee4275a0168848a1b5f0718b","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-14 05:29:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e3219f9ce9bd199a730fed2c896ae319b6692f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 08:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f0d4c41deb541abb7be690b7dda3b8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56115a94a5d489ab4dd3de47b3af268 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d48cd8939df4ea5be58cd8b23ca0370 : SUCCESS in 7m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a78f660d931400699bfd6cb589ac850 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465a56a7b6cd4ab7ba00748ea2369bd2 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/925dbe87b65e4c3a9bd3fa88074ee192 : SUCCESS in 1h 04m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb26e51ba5d44529850714bdcff8efbb : SUCCESS in 1h 06m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/596a7930e379473ba7170d9edef70ea3 : SUCCESS in 1h 16m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bdd4961f99643149388cf7e99d12746 : SUCCESS in 1h 06m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89791fe0c6bb4351a128283e86911656 : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d08401a2e70418ea8b7362d74cd188e : TIMED_OUT in 40m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/245c7add12c0445684a10b9dac3ea445 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03296ce9dc29404f86bb88eec527e637 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b4fe91eaaad4d378476acb7abf603ca : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b3c04b1a52cb4ac78f145c9c82669a06 : SUCCESS in 4m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee84937a8e8846898c3921403d545cbb : SUCCESS in 29m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51ac8e8d49db4e5481144624c7d19c45 : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dda8266fb40b4da292dfce6179293bce : SUCCESS in 2h 00m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6be2cb8b1a10423ba0707a4f69bb9aab : FAILURE in 2h 41m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc9e848d6404499997819e01d711a8a6 : SUCCESS in 1h 51m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/76d366fe6f8347169c0fa42a0c604ead : FAILURE in 10m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f97dd96425eb4bb6999834548bbf1a40 : SUCCESS in 23m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/26f74807a2d94b4c848df9b8e7e17555 : SUCCESS in 1h 15m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b1c674fa08164008bbb1f5d71c177dc6 : SUCCESS in 1h 47m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f425391d9e7d4a6bba4e2dbebda41836 : SUCCESS in 35m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/04afc92376d242c9bf6aadd1b4688b17 : SUCCESS in 1h 01m 42s\n- devstack-plugin-nfs-tempest-full-fips-py311 https://zuul.opendev.org/t/openstack/build/7b4f401073b84478b3f55ac5d4a08cce : FAILURE in 9m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a349deeb5b154e34b6883ba82fe2d918 : SUCCESS in 45m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/391101a4419b42a3b6125f220c1511dd : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f154fef5ee1c4b08325e74619d09abac069d83e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-14 08:31:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_05_2026_08_24_29\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"94459c30129a7f7383a1fe168f938e2fba84ceb7","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-14 08:41:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10f406943fced428baef9e3e5f71819c35f77019","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-16 17:53:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ae1b7f11c50d48b6b4cd868ecbc36e1a : FAILURE in 28m 47s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988252/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988252/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"2a15708668efc366a5dd5bbc21fee823c75d2a86","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-22 15:26:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bf59040735396286cf0120304400ef4e4bc241c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-28 14:55:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39cf32c4cdb4fd1a1ba7be3fed2d6921e6097320","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:20.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"7289de449e403552318b21f8dd7de92bc6970443","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-16 19:23:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5a0b708ab264bac4b7a8335ecdcf2291125fb25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 19:24:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e8d7396a8d30a7f47b38e4c834e0d1064eba7e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 21:24:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e89301ab1c488982918fad8810910f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87d6cfdaf7f4ddd99cffce7d3307f15 : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/021de6bd79d14488a71b0d2021864a74 : SUCCESS in 11m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9849e8930c247ecba7c3939abb43df1 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcefada473d346459c63af6f73fb7c86 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c014f5f345bc4492bbda620bd6e8fa96 : SUCCESS in 45m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e52774a3e2664735877b47434ee2d403 : SUCCESS in 54m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e1aff82e48744708dcfe8efe7de8de3 : SUCCESS in 1h 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3799751686c74a00a3a33e1e23f052f3 : SUCCESS in 1h 07m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/188a49b8029743cabf4031728a1f5dea : SUCCESS in 9m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1cce228a9e66431089eaf685854185de : SUCCESS in 1h 25m 22s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1d2c7fe00dd4f86b665fba6d4f72922 : SUCCESS in 1h 55m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/756e611cf3664f8c9a2bbd6dcf9de62d : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e690d9f0ea26e90bf9dd0038c086c5abd608756e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 21:24:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2f75901c6060bfa6f53b44ff70dc665e0a288eef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 21:25:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db51d576792e438daad03f90db1e5adf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218df0ece0d04e1abcd1332a3df71b55 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44cec43708e740e1916d9ed263fda677 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"4d8200dac208c309eaae18c77defb65e5e9f4b62","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-17 12:09:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-52-988252-1-fc : FAILURE in 13m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-52-988252-1-iscsi : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37aec4236a4e74b866cc71b80424300e679ec90c","revisions":{"37aec4236a4e74b866cc71b80424300e679ec90c":{"kind":"REWORK","_number":1,"created":"2026-05-12 12:49:49.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/52/988252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/52/988252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/988252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/988252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/52/988252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/52/988252/1"}}},"commit":{"parents":[{"commit":"e83f4cba8722dc06ab7396ce2239a272b0345da8","subject":"HPE: Skip UT that fails intermittently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e83f4cba8722dc06ab7396ce2239a272b0345da8"}]}],"author":{"name":"Alejandro Santoyo","email":"alejandro.santoyo@canonical.com","date":"2026-03-16 11:55:23.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-05-12 12:49:49.000000000","tz":0},"subject":"Fix properties\u003d{} being sent to Glance","message":"Fix properties\u003d{} being sent to Glance\n\nThe changes to fix LP#1527324 and LP#1823445 have as side effect that\n\u0027properties\u0027: {} may be sent to Glance when creating an image from a\nvolume which fails. This change ensures that the code in\n_translate_to_glance() effectively removes the \u0027properties\u0027 key from\nthe metadata dict if it is empty.\n\nCloses-Bug: #2144550\nChange-Id: Ifd632433ce5759494172b74ce311816e10728d21\nSigned-off-by: Alejandro Santoyo \u003calejandro.santoyo@canonical.com\u003e\n(cherry picked from commit 2096a784695d5b7304822693b595d9238b66209d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37aec4236a4e74b866cc71b80424300e679ec90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/37aec4236a4e74b866cc71b80424300e679ec90c"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"ours":"e83f4cba8722dc06ab7396ce2239a272b0345da8","theirs":"2096a784695d5b7304822693b595d9238b66209d","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
