)]}'
{"id":"openstack%2Fcinder~988315","triplet_id":"openstack%2Fcinder~master~I4941197a2df741f80c317fb9e270f29f6fb085b2","project":"openstack/cinder","branch":"master","topic":"per-pool-image-cache-promotion","attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-08-28 05:08:55.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-08-25 10:59:12.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-06-09 16:15:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"39225":{"account":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"last_update":"2026-08-12 10:02:52.000000000","reason":"Someone else replied on the change"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-02 03:52:12.000000000","reason":"Someone else replied on the change"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-25 04:41:06.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2026-07-08 16:58:09.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-01 12:01:30.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-06-09 16:15:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}}},"removed_from_attention_set":{"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-09-04 07:04:56.000000000","reason":"\u003cGERRIT_ACCOUNT_36179\u003e replied on the change","reason_account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2026-08-28 09:47:24.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4941197a2df741f80c317fb9e270f29f6fb085b2","subject":"Implement per-pool image-volume cache promotion","status":"NEW","created":"2026-05-12 17:56:10.000000000","updated":"2026-09-04 11:04:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":120,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"98da6c889bdefe23424fb72f822827b540e5a0e7","_number":988315,"virtual_id_number":988315,"owner":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 05:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-12 18:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 18:54:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 20:26:44.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-12 21:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 22:21:06.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-13 00:10:15.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-13 17:56:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-14 00:11:23.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-14 00:22:09.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-17 00:43:39.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-20 20:35:03.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2026-06-09 16:15:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-09 16:15:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-06-09 16:15:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-09 16:15:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-06-09 16:15:28.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-19 20:01:40.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-07-08 17:45:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-08 17:51:39.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-08 18:57:24.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-08 19:06:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 13:16:03.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-16 13:45:37.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-17 10:23:29.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-21 05:10:35.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-08-07 16:02:03.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-08-08 07:21:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-12 09:40:13.000000000","updated_by":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"reviewer":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"state":"CC"},{"updated":"2026-08-21 15:51:43.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-08-25 05:57:52.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-26 14:47:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"}],"messages":[{"id":"3fc7595275af818391ed7d35e5a6f81144f52c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-12 17:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"370dae06a01e1b2321459257559e42a5e67a8779","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:19:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a949f127d30941fd9944ea0ee1e802fd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62bb051091ef4d539e365e1c8bd5ae5e : FAILURE in 6m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0ed52f0f5f8042598f1f48480716f3ed : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c944a0212c0c4326a9ee383e20db88de : SUCCESS in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ed681195dc6745b977a07a6bb5818528862d3c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 18:54:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":1},{"id":"0015cb9d3c2a2771e053ef6cf1d36a5b8fab251b","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-12 20:26:44.000000000","message":"Patch Set 1:\n\nBuild Failed: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/988315 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ed59c541f539c58a29215085cfedf1694ce0096e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 21:12:23.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73318ef9cb0a4278807e66ccabc36492\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59325924c3944d77b649a27f973a48cb : FAILURE in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c58135eb79dd4bd7852ea4cf66554a83 : SUCCESS in 11m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3aa282a97e9c4ce2a38b472bd56e3ba9 : SUCCESS in 9m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c65e5a91b124019bad3937ec97afae2 : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9426b7f12b2e44dfaa5851e936d05fad : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/88dba49cfada4a36aa10bca15ea7e323 : SUCCESS in 30m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a0b3fa850bb04eb5a91db9592159e5dc : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7cc6fbaa277e42ef8e20a017bd5fa89b : SUCCESS in 1h 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9041e274c38249de9fa0239311c531c0 : SUCCESS in 1h 09m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b7060632d6a40b2bd6920ed6e760e4f : SUCCESS in 7m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb7eea7aa167470cbbde644b05782930 : SUCCESS in 17m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9241de373100486aaca8564b9103ff67 : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/537b2b36ea7f4099b60804aab6f88847 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8799e90ac33d46fb80d1eb28a6de8df3 : SUCCESS in 4m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/010a2617c4f346fd87c2b88f12c39f11 : SUCCESS in 5m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bd3ca781a7b649aea4ffcfe1ae05d7f6 : SUCCESS in 7m 20s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/246ffe08cbe9460fba96ffdc40e4508b : SUCCESS in 38m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8de75e9bfb643e5b41ab29aab3033dd : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/469d9c0c8632491b978dc43a6c6b91e8 : SUCCESS in 1h 52m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb192a53bb4644d0bbba2d5150a2b2ee : FAILURE in 1h 33m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c5dd9a961cf1484683d2feab3e2a33d1 : SUCCESS in 2h 16m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e090f243c4dc4f1c9f63872820fe9e74 : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/666d208095f24e26987c2e09ac028d54 : SUCCESS in 25m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/87ed7a40bf744e14b5acae02a625a158 : SUCCESS in 50m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb2aa9a10a5046b8814dd41ac6603685 : SUCCESS in 2h 07m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5728e88da0448c992d5bd164b26e57b : SUCCESS in 23m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/26c93863eff542fb98c14c2654ef929c : TIMED_OUT in 3h 09m 56s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ac95345436bd47079ac33878a87e11d3 : FAILURE in 1h 30m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a33731c99cb44a1c8fc1c926ac0fc482 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0d55aa5bdfd12c5fcf056c25d636717c8c3e3abe","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-12 22:21:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/e63290715eb745b3945fc2addee86b13 : FAILURE in 1h 14m 45s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988315/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"eedc72a6c2a5e4863fc22e1557b6ac4f45f1c698","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-13 00:10:15.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988315 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"5400ce45d64cb395e2fe91116f34fd3dc09ba680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-13 16:31:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"576a22d0ce766bebcba18bd7d4af4198ee2324e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-13 16:57:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41944be3db8fbb332d8a4752692f72630d80d28e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-05-13 17:24:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"562f546d7f4129951b0e183c3f6dab6e26c6ab9a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-13 17:30:09.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":2},{"id":"211328baa545d0cbfe0f44c7f8c125cd0024f95b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:52:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f7804bc10b48e08d290554454ffd84\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/688110b43556465f869bdcc685790556 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea7a523bfbfc4f608278c44e3b3643c8 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3bc61dd652584d3b8f6ff01b1768bfb0 : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"85d5c98d1ec88e70fbc9b91d945e573f73a3d541","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-13 17:54:04.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":3},{"id":"41e4db32bc11d12004b44e5a6b84cc3d246563d1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 17:56:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4131/.","accounts_in_message":[],"_revision_number":2},{"id":"b42b7174cf852e8c54970961606dd98afa0e3c92","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-13 17:56:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver build/9d398591dc894b53a731a6ddbe4d0cf6 : FAILURE in 30m 46s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988315/etc/tempest_log.log","accounts_in_message":[],"_revision_number":4},{"id":"0207e141476fa705e730b182c06100da06276bf0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 18:00:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3955/.","accounts_in_message":[],"_revision_number":2},{"id":"834d70d2a4631b9b5f2dcfea49ce484d24dc9b97","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-13 18:11:42.000000000","message":"Patch Set 4:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988315 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c056ef581cb55642a6776e815872be4be401d560","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-13 18:21:51.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":4},{"id":"fdf87429bf735975ed45b953999009a71fbd3015","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 19:04:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4132/.","accounts_in_message":[],"_revision_number":3},{"id":"8921b6ec0e6ebe808e7d6f8a30d17b82127077a4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 19:26:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3956/.","accounts_in_message":[],"_revision_number":3},{"id":"2ccd8c96f6207f999473fb205484eff9a5316cfc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 19:43:37.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4133/.","accounts_in_message":[],"_revision_number":4},{"id":"86cb201a72461dd3f1e316f0fab75cbe9452e07c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-13 21:01:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3957/.","accounts_in_message":[],"_revision_number":4},{"id":"e522cea92a0147f44f84cf0e40404663caf19d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 00:11:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce38df5ba08a4c539d252cd1c4c77f9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb0c39177ea48e68c1f5dd3b83d8020 : SUCCESS in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0ebdca8446540e4ac43dd437ee2e1a2 : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/726d6c4da8c94cda959ca9640b470b74 : SUCCESS in 9m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eeb2902a294c4e79b76f483b5165d53c : TIMED_OUT in 40m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0ecfad0ab944418eefea512640a7ae : SUCCESS in 14m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/b03128399aec47998fd6f00144f43510 : SUCCESS in 1h 15m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a186b0260c4a41f88349e0e3b643215c : SUCCESS in 1h 06m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a134c5b20f943afb0439eb5111af560 : SUCCESS in 1h 58m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7afc81150cc4459f8b8b3cbdfb7b3dc5 : SUCCESS in 1h 15m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79cae0cf2bae4b6fa8176c8a98bc5034 : SUCCESS in 8m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a2714dadc3e94e7bbb70712cd0f82e8a : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e817ddd4b01a48b1986c455230c2be6e : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45b15c725a274b08a106a368a6775f16 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/de0b5268b5614c29b29486a44f8dbda1 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/24f98b03f2654cb986edbc44dcf4a0b5 : SUCCESS in 5m 58s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b95afe2253354972b1a4517d44244487 : SUCCESS in 8m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/efec4b6996bf425c8af9ecdc6d343c9b : SUCCESS in 41m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a7618990022b41a4810686c994c27d80 : FAILURE in 3m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2f4799252fa54fc4a3a048078da40028 : SUCCESS in 1h 20m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eb36c8274243490d9f8b3099d5801eef : FAILURE in 2h 54m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f7a91607ad54349acf067e9b930c95d : SUCCESS in 2h 01m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3a42534f17b84d3c980a25d831c5a88d : SUCCESS in 2h 05m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a0b62113cf354a1188172867ab03bf55 : SUCCESS in 42m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/843eb02b8ca04678a3e21e846b9f9cf3 : SUCCESS in 54m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/deebee9d1ed0452ea331e07bd018c952 : SUCCESS in 1h 55m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3fc06524dc464aae8172dcc4935c1cf5 : SUCCESS in 33m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/00d1d120aa8246cd8ce0949226e756bf : SUCCESS in 1h 05m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/554c254e4706422186c44237e32aa0f2 : SUCCESS in 1h 38m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e731066db421461d973175fbae5a8194 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":4},{"id":"6e16cd0bc9e538d367544f508d3811ae42c8ab48","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-14 00:20:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_05_2026_00_14_33\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"eacb359e8c29d1ec525d12f5fd76d46db3e8f11b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-14 00:22:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-4-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-4-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"1c48131f0a78734b7fd59f7ea605a11c7ed9822a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-17 00:43:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/25465266453742c8a55a76ad134dfea4 : FAILURE in 28m 17s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988315/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988315/etc/tempest_log.log","accounts_in_message":[],"_revision_number":4},{"id":"cb186bf721633f7669f381c8a71a055fc2841d46","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2026-05-20 20:35:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e2b1f0e406dbb3450299785ae8982daf13433b6","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-09 14:58:16.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0fc49719903d248a34c709adb2f481b46481ee98","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2026-06-10 14:16:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7a99b4a030680d067dc32c841982ec517d6bd1d","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-11 13:33:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a994e540af013c0ad53e0a07d01e897b504d7292","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:40.000000000","message":"Patch Set 4:\n\nBuild Failed","accounts_in_message":[],"_revision_number":4},{"id":"fe09cad77a06d9572db93b2bd9a906832a245608","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:04:30.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"086747b2aecb803516a6d084d3c3eb4776c12851","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:34.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"6fcaed5bc00e3610f53865c069647737af16a6b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:06:00.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3},{"id":"ada2701a9bb8386f70629eb3d00c9f1ba8d45a6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-08 16:58:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6e56e6aef2a9522c6490f29132308ebb295fb01e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-08 17:03:06.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"22e5ad41639db559394fc9d77533a17f1e75ab8c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-08 17:45:30.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/fb3efbe91e3947df9b8fb3bb4ef38a99/","accounts_in_message":[],"_revision_number":5},{"id":"e6bbb8f782f793517841fbe7a9d32705cf68a0b6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 17:51:39.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/540 : SUCCESS in 0h 42m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":5},{"id":"11bb52ef5d06f93c3d5a7ff0abb90dff6576673e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 17:55:14.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/455 : SUCCESS in 0h 43m 41s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":5},{"id":"eaecef654a63c59df5fa49c87217464e1d71dc7f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-08 17:56:20.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/811 : SUCCESS in 0h 43m 1s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":5},{"id":"4fd9d23f8af7440da482db4871d93b0c845d548d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-08 17:57:27.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fb3efbe91e3947df9b8fb3bb4ef38a99\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ecc8f5607a3462c9036abd8eeedde95 : SUCCESS in 58m 47s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f21387b4b77a49b6a968eeed45703245 : NODE_FAILURE Node request 200-0000315722 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7833607743b94258888725cdf153fe76 : SUCCESS in 46m 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7eb6fc7543e94dc7b5921415b398419f : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":5},{"id":"872ebd936fb0b6b58d8cfa3b49a11ca149c810be","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-08 17:57:28.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":5},{"id":"4be48bfe9b0c3e425fa97faf2cba8e8412932654","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-08 18:57:24.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-5/FC/ : SUCCESS in 1h 58m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-5/iSCSI/ : SUCCESS in 1h 42m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"010c02ef3ae327e18feaed62a76b867c7fafa2b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-08 19:06:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/8ed37ae8fd2b4cc89270e9c22de8df4c : SUCCESS in 48m 01s\n- cinder-storpool-tempest-iscsi-multipath build/6bba060e73ad43dcb9cfb413665ee07b : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":5},{"id":"392ae1400dae8a254eab6ff8bfa2a8158dd7b858","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-08 19:08:08.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4731/.","accounts_in_message":[],"_revision_number":5},{"id":"a46c145837e22f37e1e9ff8800c4b767f6d37191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 20:51:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/561c38dddec64847b9024f205663043a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/624e3e9eb15a46b2b31cd70adddd1663 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/879445c6751445d992f89212045550f1 : SUCCESS in 9m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/910b9a888d08422ca92f9e8d12c8f101 : SUCCESS in 7m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94fa49b92f9f408fb1ef75a4325f466b : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29533aa030843b3a88e3ba67f8a83e5 : SUCCESS in 15m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/e5bbc21fba6b4e5191ad9b5fb8e84b47 : SUCCESS in 1h 11m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cff7bb697b634c059ce2e1dd731fdb79 : SUCCESS in 50m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a34713ad356497b8ac04381df29df6c : SUCCESS in 1h 25m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/352393b5c2974e419551ca1c8a26e239 : SUCCESS in 1h 07m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/623d0cb1865f4f9c95b9102a2ed35895 : SUCCESS in 8m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1904805087eb489db08d00b7abdd76fc : SUCCESS in 28m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/562573891b544084af443d83661e2e07 : FAILURE in 3m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea682e381a3f41eb9986e3629cde7af7 : SUCCESS in 6m 24s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4668058612364e23bfe97766fb379af5 : FAILURE in 14m 12s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0219634f702a4bdb92b37adeba62cb82 : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/002aef1dfff9455587c93cfefafdcb9f : SUCCESS in 6m 59s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7cf750d8dbf749368b0e53a67faece39 : SUCCESS in 7m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7692a2e651584f98aa9c7e6d6cf780c6 : SUCCESS in 42m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8afd98e3a5934eacb933c63a2c4e10f1 : SUCCESS in 3m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd153fd73cb44bb0b3c579971bf46b2d : SUCCESS in 1h 56m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/968d4284c1f44dd79fdf674002b0fe88 : TIMED_OUT Host unreachable in 3h 46m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10ed5778c74e45be9b5e38193b754dab : SUCCESS in 1h 34m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d3875b42631a432ba4f737a3f31ae2d7 : FAILURE in 3h 16m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f32a21a52f5c43cc95bf390216e5d640 : SUCCESS in 2h 23m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c4ac1c16f6634536a46a0d70b98830ad : SUCCESS in 40m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bde29e43e5b7418188669bac3f090f94 : SUCCESS in 1h 23m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a892b609416347edbd42b55a29170b11 : SUCCESS in 1h 52m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6906a7213b184bda93bff6cf2796af9d : SUCCESS in 50m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8fd3279f98a04af6bf2df061e1bab837 : SUCCESS in 1h 28m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b79175f7cea14d24b3ed6b6ae2c0acac : SUCCESS in 53m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d470bba354b4e41beaa47ac5623f201 : SUCCESS in 1h 05m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"6c9dca21c2da54c0349ee4015bc1d0f30ab2c8fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 21:01:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b9ad4682474767981cafc4658b54d2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acd5cd3f71254034a7199c2056a30541 : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/993c76a665be4257b66fa1c7bc130e6f : SUCCESS in 21m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/acb6c2e549804f6591d355cac29e12c1 : SUCCESS in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1f3bfa4c8984d5d721a5197f7c698a49ed1fecb1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-09 15:55:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-5-fc : SUCCESS in 1h 18m 23s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8a19257a6c0887ec090b6f005d249e2165d85617","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-10 07:38:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"85ed8c55d28159c41729768760101ab8777a0b49","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 08:25:35.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/dec2f281091043148d83f19ef4b52e69/","accounts_in_message":[],"_revision_number":5},{"id":"6d7c64851480195085a3ecdbf756632d52d94999","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 08:54:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dec2f281091043148d83f19ef4b52e69\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0296d2ef01ce4034b3ff788ebf4a85bf : SUCCESS in 46m 48s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87ae4fa25faf4afc9fed13a239e263a1 : SUCCESS in 40m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ab14585bae849d19fb91a87e7ae7713 : SUCCESS in 43m 37s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2781d627d7234b9a8e759110bd935f29 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":5},{"id":"bb0bc4c8484d8d8050302f52538c937d2cf189b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:41:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1557643427e44c11a93348a847801236\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0073981ef0a847ff90daf8e9f5a50b97 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65786aee37f442eca373a76a82b3fccf : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b50c6ec249b941cba7a7283bef13314a : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7061ebfe7d6818dfad97eef065e0404bd2b7af9a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-10 10:00:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/48f04d7381c8481e95c65bb00fe49320 : SUCCESS in 48m 28s\n- cinder-storpool-tempest-iscsi-multipath build/241e4009f094401db3401e5281075b84 : FAILURE in 1h 01m 22s","accounts_in_message":[],"_revision_number":5},{"id":"6fd0c422603ba0f48b61157c2c85e489458fdc11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 10:32:25.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbf4325e83b64bc6a52754509694ac75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/475198fc1ddc4f9f95948799583bc579 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0316ba6c09ac42798eefed29968865d1 : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d6e8c44602e47b8a157eb340ffb8293 : SUCCESS in 7m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0b13c8f6fcc41d89df90ce5e2894e50 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ce758310664147b1b3394b27f3dca1 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a28c61a7b33b43e39f53677c886f7fe0 : SUCCESS in 1h 05m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa91f71ce6584813878213c65e5c79a8 : SUCCESS in 49m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e9e23eea1e5146319b66752c61526e5e : SUCCESS in 1h 28m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c074ab9ca4d4ea082dbb9780d52e817 : SUCCESS in 1h 08m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f413f362eb54838846a132aed752ef6 : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/58c22892656e4050a5252ce5958190c0 : TIMED_OUT in 41m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/72649b4e4f624fa9b6281af9d097ffaa : FAILURE in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/446053648fca4ce1ae838a260581fd9c : SUCCESS in 6m 53s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/9e99f87748ea46bca1a65e5598ac0288 : FAILURE in 14m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbaced801abe4fa39cc86510ad2aa55f : SUCCESS in 6m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/19456961af314c949054f58e6a215a33 : SUCCESS in 6m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f64badc550954c38a41eb97426564351 : SUCCESS in 8m 41s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c109ae75dca4b21848424fe04166dd5 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48cedc38174a412abbc61c9179c934bd : SUCCESS in 5m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/62d93d1c8e8244f78d4ea44fa91b0ba5 : SUCCESS in 1h 56m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6bd4fccdd9da49a7a5b4327fce8b3e8b : FAILURE in 2h 18m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d3e3d3cb57348d1991bb52acaf44e49 : SUCCESS in 2h 13m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/27f09016fd53437cac9aa9e64efa07a7 : FAILURE in 2h 45m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1652eb174f304459940ce4350c55b1a7 : SUCCESS in 2h 19m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aee287c3fd6f424cb03fd997f63eebde : SUCCESS in 37m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ee7f58bb1b64dc2ba97c199ad02e805 : SUCCESS in 1h 21m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c9649a6a1a9f48869738487224b57abb : SUCCESS in 1h 58m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fa1d54e2d7ed4254a1d8b953942e7a37 : SUCCESS in 48m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dc6c9b7c7aa745359c70aac5dc5e50c6 : SUCCESS in 2h 05m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3fffbf694346485492707ca6b61bf44d : SUCCESS in 1h 51m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6fee9afb284e8eb327c669e7c79857 : SUCCESS in 1h 00m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"d7c6fb51d09ad43ea6c08b589a4fa296eab0c56a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-10 14:34:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8bd3c0f65a3976f8f6a1c7a2b94413481ed6034a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 15:12:15.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3c47cd08bd654690aeca4f34e82b8ccf/","accounts_in_message":[],"_revision_number":5},{"id":"679d1f14933d653ce047d06ef301503b42a6b8cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 15:37:03.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3c47cd08bd654690aeca4f34e82b8ccf\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/619bddd81c5a4d729ad5f5861e98c3f4 : FAILURE in 1h 01m 56s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/033f74b619d84b85bf830a522fbdb50b : SUCCESS in 37m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f109833f7c94694a7511aa7b0516a31 : SUCCESS in 42m 34s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/753da201b9854ab2bcd2cac7b7e6175c : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":5},{"id":"181833abe33d4a86e81e361105358d0896cb8409","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-10 17:35:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f245e292e2204ebfa1c0eb5a14ffefeb : SUCCESS in 48m 24s\n- cinder-storpool-tempest-iscsi-multipath build/bd3d886760ab489ab6d1630e85e71401 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":5},{"id":"f9b048522b061017df26a2f3765902ba77d53c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 18:01:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f89f41fb332246d585e017d18e6eee04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89336cd246e4d3bb8139bf03a7cb8c0 : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edc7192bc8c3470a8618a22406c1ab16 : SUCCESS in 14m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a474b9100604b31a55da6262623dd34 : SUCCESS in 8m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e586872bf90645e9b0f4243c3a466a0d : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a02011657d748e89288ae1dcfd7c1c0 : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/8eb9ee553f424ec4947892404de35dac : SUCCESS in 1h 03m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c0cb826a97345efb9d21423d13cbc1b : SUCCESS in 58m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ec3a594dc824adfb804a4bcbf0b12e8 : SUCCESS in 1h 48m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03385f0118254a5bbdd4aafac6cf70ff : SUCCESS in 1h 08m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d1020d7b1b145c5a696b270250e91c6 : SUCCESS in 10m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d7121da9aba74039bcc818fc84dd06f1 : TIMED_OUT in 41m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/89b08a2030a64794ae8888113473acb1 : FAILURE in 3m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e309c2de1c34845807fa2a0f6a12b23 : SUCCESS in 3m 41s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f7ba18f833884a37baeb87919431ddf5 : FAILURE in 14m 05s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/363a2e12ca9b4e6a88eab7565aba69c4 : SUCCESS in 8m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6d4b995bf8d14b3788bb7f7c277b2255 : SUCCESS in 5m 56s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c54d2701415b490f96c456c20f4e90f9 : SUCCESS in 12m 43s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a16e9dceed1a4b618c648ea55b22ebc0 : SUCCESS in 41m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e0e2d82701cb43f4afca5e76f19d3501 : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/aa1bb38bbd1149eaa2f25b713b5333fd : SUCCESS in 1h 50m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d840cce4a934589b5720c9735cd14ff : SUCCESS in 2h 48m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6f52a3b5bf204f369a75acd764e9720f : SUCCESS in 2h 12m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8ec94d282dc4431eb4c58bec31f56335 : TIMED_OUT in 3h 20m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/47f5e8ae90e549c1a095b4bd9e4da488 : SUCCESS in 1h 52m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c3c4e3e704c04c098ed2f35677395531 : SUCCESS in 38m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5c8526a6733c4378a66bf64aca372cb8 : SUCCESS in 37m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bef71fb0ca2e4dc4af87060bdf401c8d : SUCCESS in 1h 58m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5e4145470fb6485fa731abb000b27d3e : SUCCESS in 53m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cf170f63033641938caa4be8d0fd1596 : TIMED_OUT in 3h 08m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/294aba058beb4cf9beaafcde72e5d0d8 : SUCCESS in 1h 17m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dda8852a96246e4978b6d8b86677b7a : SUCCESS in 59m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"8c879f96d5f4feb3dc178a5cf71af57564b5f6cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 19:23:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a675228b3bf46adabcb2f2547e281fd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5b61e22d0884d9aa78315b2e1366c7b : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/11bd1b177b8e4b2eb2d979a52214a0d8 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d6341b9b3d89420395c272888508749b : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d983c0a15bd272dec0464a4b8646de6a45e88bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-16 12:56:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5217120a0e07e55d4d760bb2f846ccc42a2ce37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-16 12:58:45.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3e35c91e540e5ab8ce04164ab81bc27092c57bac","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-16 13:16:03.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/624 : FAILURE in 0h 14m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":7},{"id":"3777e5e2add9d18d80490c4b5e41598f9624a402","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:17:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79728d72471f4a88b26ef2197b84bd7f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/046d32d9ee6246f99145a7950de83dd6 : NODE_FAILURE Node(set) request c9c59c576c0f4c308edc7f797d4b12d7 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86d1583e9cee4f39a5f4492e5fee19aa : NODE_FAILURE Node(set) request 069e1458f8cf4b4195065347a25a48c0 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1234580d679046b8864e6b35742caf6f : NODE_FAILURE Node(set) request 7591fb85743b4fb8baf1e582e1a22b8e failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"165f37f3583e97f1d60c3f472d8072b3a8a394a0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 13:24:50.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4810/.","accounts_in_message":[],"_revision_number":7},{"id":"ddfeea36f47cadf2d5226d29f4552a67f173f09e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 13:37:17.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/73af98406a92432e8a5dcd690b8e9fc5/","accounts_in_message":[],"_revision_number":7},{"id":"4d5c20f24e33285442b56751975c486a50295fa7","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-16 13:45:37.000000000","message":"Patch Set 7:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/532 : SUCCESS in 0h 45m 0s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":7},{"id":"3c23185d4bda485ad7713a1c797712ad6c2cd320","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-16 13:46:45.000000000","message":"Patch Set 7:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/887 : SUCCESS in 0h 46m 10s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":7},{"id":"5066135c76e06d4fbfa9a94262d9149c469c1dac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 13:50:39.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":6},{"id":"1b3906b15c0360e90a299e720d52fc3f26e7e55f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 13:51:59.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":7},{"id":"ada16afc6d71b9c0a3afc994c606a2aa1a5d4e4d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-16 13:54:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5af53d7fc3b649358a230ed7e58694e9 : SUCCESS in 43m 35s\n- cinder-storpool-tempest-iscsi-multipath build/35281b2007a54054bd4ee6ccbf8af41a : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":7},{"id":"9eb1bd008e07922f1bd12ffb10aabfee123abc1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 15:14:36.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/73af98406a92432e8a5dcd690b8e9fc5\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74d1e1980984480095e6944319117b7c : FAILURE in 41m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a30eb587a5d54759a42656a1c2eddb96 : SUCCESS in 35m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cc04ee2b08f0475cb3ee778d68d39e44 : FAILURE in 44m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd052372054549d5b32ed453b062c4a4 : FAILURE in 34m 04s","accounts_in_message":[],"_revision_number":7},{"id":"45b20da29a173ad7a637b2b04969396ba11ef455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 16:13:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/376e7875cd274f25a35b4e5dfe75244b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef16f9bbab445179d94e38db283c65c : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b78642626d4455485c3338397a14b99 : SUCCESS in 13m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/406fe30a95e14921a03e9a5cae81a128 : SUCCESS in 10m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/952550ed08c84631a4aaaa503a0d1dc1 : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b9e926c9d04c16aae61ef3a3860bad : SUCCESS in 16m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/bce877de56844644bc35eef895905770 : SUCCESS in 1h 06m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5192611f974a43e8b10c1c137c4fee3b : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9407bb68c06492a80bd219682a26886 : SUCCESS in 1h 16m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e28a849103da4d2ba3aa770f465b9323 : SUCCESS in 45m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc29b22d3be748d890da95d9d76f3e5b : SUCCESS in 11m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7031a671e0c449b8813ceee40306fb1 : SUCCESS in 20m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2160cbe05b3749139751d98d4e4048a7 : SUCCESS in 3m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4ab8c943fc44680804dddf0874ea09c : SUCCESS in 7m 29s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/32262bbfd0a04667b0eb347ba346330d : FAILURE in 13m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ca33c7e6f7cf4329981843d92fd55c6a : SUCCESS in 5m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f9348019074d431b8f9002c49805cbbe : SUCCESS in 5m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9c37181fd01d42f1be8622a6d1b88954 : SUCCESS in 8m 57s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/390d8249ca5c45028f6c7e26d45291dc : FAILURE in 50m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a9351d8f10942c492c04d789267b20a : SUCCESS in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a6fa1949b3d44059b06f86543e5fe6a : SUCCESS in 1h 13m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85cd81f4e5a74d0da9b1378dae907154 : FAILURE in 1h 58m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a12b5e177ea448dc985a7fc09f6a4931 : SUCCESS in 1h 40m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/11d76744b3e34744a3f98c0249d9831a : FAILURE in 3h 09m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/44249b03130840cfbe1559c4ffebc263 : SUCCESS in 1h 51m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6128927d555443188b4dd8782e93dfd6 : SUCCESS in 36m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aa00651df9dc4984a912f56bfb98472c : SUCCESS in 59m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/83a5466d8a004698bc55a6c3a9f3b0e2 : SUCCESS in 1h 33m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0bdc10f73fa34f688dcd3fc651acb45a : SUCCESS in 45m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a489886ca10348f2b9c5330b53112418 : TIMED_OUT in 3h 09m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39bc1ee4660f4c58b4229eaeca002a94 : FAILURE in 1h 02m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73c24e4890f24993aaea0020583cab74 : SUCCESS in 54m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"0202cfd229c43fb8937f8bbe4b6356ee7d715266","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-17 04:54:27.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-7/FC/ : FAILURE in 15h 55m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-7/iSCSI/ : SUCCESS in 2h 40m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"374a0d228e5cd3fc4c9b095cad3bf9a97bf7d13f","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-17 06:03:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cde8dccd31dba700690f93180824f9552c111462","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-17 06:37:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9923f03b14bd424e90fcbc19aa16fbdc : RETRY_LIMIT in 16m 29s\n- cinder-storpool-tempest-iscsi-multipath build/aa71cfaa64f740a0a4cf15f6c7ee5ca7 : RETRY_LIMIT in 17m 37s","accounts_in_message":[],"_revision_number":7},{"id":"7c442e518420897cdd3cf00172a3d72aa0850b50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 07:06:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12b7a8a36f774ce094d54bd9909fa106\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55928610fbc94d70b369bc4b54a63622 : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fda20691485f419fa4175e599d108121 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/72b6c3f5863347a4a498b82efefaf866 : TIMED_OUT in 1h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20a3fa898b6a05b8bda59c5788d5ff026d86a908","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-17 08:02:38.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-6/FC/ : FAILURE in 19h 5m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-6/iSCSI/ : UNSTABLE in 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"b0843827c7055ff8afa1037dd743d839d0b9b425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 09:41:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14dd951611624105b9d251b4c85ec44c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41880f74d7af410a9de3f229be3e36be : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b78dc0c7473646da8105d2e8c33856c8 : SUCCESS in 10m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d37017d983f24a29a81d2cfd30dbd79b : SUCCESS in 7m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/338c253452914136a4efde78674595d6 : TIMED_OUT in 40m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371c49988e4042e8833f2721b274887b : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/fb2c67f412834e23bf45efef2d73a76a : SUCCESS in 1h 01m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea7f447575924dc18d6eed25685517cc : SUCCESS in 52m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a52d107837549bdaf9e03111df79f00 : SUCCESS in 52m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24a02073dae44f518abbc2732f5ef60f : SUCCESS in 58m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a691c56e07fc4cd2b5639aacd22d7d12 : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/11bd4d589c5a4656a5f8b0b71008cf33 : SUCCESS in 24m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/844e5e9766c847738bdd084154bed235 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ea06050a97a4764a52e9503c9e11873 : SUCCESS in 8m 01s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/4bb6b1b69c304ca5932c420017fc4164 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dfd5a678bff043bb98371902def9ec1b : SUCCESS in 6m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/35efda458a0e4fd495bb43b24585891b : SUCCESS in 6m 08s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/14679d0de1474ae199fb1e6fd52bd117 : SUCCESS in 9m 32s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/49e09133e7e44bbaae90b66cec696822 : FAILURE in 47m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48ff5697ec5c4259956bd8cf9241a2b7 : SUCCESS in 7m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17919950f2924d8d80ca394720d0bfa2 : SUCCESS in 1h 28m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/144473b2283b4e5ca9faf9593f17d8e6 : FAILURE in 1h 04m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f38b18c4684349438c42f36ec456fed7 : SUCCESS in 1h 29m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c42a72e5466d4bdcbb625f4263689079 : TIMED_OUT in 3h 31m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/98fef7982ded4221975e0b912db0af94 : SUCCESS in 1h 46m 48s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/24790694ed2247088c205508a91466b2 : SUCCESS in 36m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ffa22fe608384657a3025bf331a61d52 : SUCCESS in 1h 10m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/925728f468314b9e8feacf5c71875da7 : SUCCESS in 1h 24m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8666a3bec5c4da58db9350997033887 : SUCCESS in 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b23bb39d5174b3993866d1dbc68b4f2 : SUCCESS in 1h 43m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ebac62afd6f450084cc19bdd779b6c7 : SUCCESS in 1h 58m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95527d12a5974b228dc2bb1fc04e40b9 : SUCCESS in 53m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"c7830b224e573bf6bdce9a4640a911ed555bda75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-17 09:52:04.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_07_2026_09_45_09\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"f3559a6748c64935fbe319d99b65e679dde34bc4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-17 09:52:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b027385a6afd6d02394e568d901f2170e5523ddd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-17 10:23:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/7/upstream-check/cinder-tempest-plugin-ontap-fcp/10bd382/ : FAILURE in 25m 14s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/dc3bb37/ : SUCCESS in 26m 40s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8ace28f/ : SUCCESS in 29m 34s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6b21fe6d1f40fb26314e1eac538a0df50466281b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-17 16:21:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-7-fc : FAILURE in 10m 43s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-7-iscsi : SUCCESS in 58m 1s","accounts_in_message":[],"_revision_number":7},{"id":"0600f5c8f04d0b76f653c54efbf38c6308b45e57","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 08:16:04.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4633/.","accounts_in_message":[],"_revision_number":6},{"id":"a0af6b303c06dc2626a9f4a6ae9ed41188cb6094","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 08:28:40.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4634/.","accounts_in_message":[],"_revision_number":7},{"id":"fe95681406b2339448a8c6259e21c185426eb010","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-20 16:45:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-6-fc : FAILURE in 11m 51s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-6-iscsi : SUCCESS in 1h 4m 55s","accounts_in_message":[],"_revision_number":6},{"id":"30bd69f37e95c8c41312d4b254901f89673f98d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-21 04:39:05.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"33067d39d1e4ebbf723facf61c9309f0a68dd7b7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 04:48:01.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4684/.","accounts_in_message":[],"_revision_number":8},{"id":"2761cf9e1148446ac2efcbe4784d2275fdbb6e89","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-21 04:59:03.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/62fed4b8bfea4ab28c6e6411d3fc02d9/","accounts_in_message":[],"_revision_number":8},{"id":"03cf5b87f310a23658af3c7f87043d2643c36a42","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-21 05:10:35.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/988315 : FAILURE in 30m 09s","accounts_in_message":[],"_revision_number":8},{"id":"12db251aa4734393bfb69263d6e015ca97c8a352","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-21 05:31:17.000000000","message":"Patch Set 8:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/921 : SUCCESS in 0h 51m 47s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":8},{"id":"b4be6a98ed7b04543e3f057956bd4b10886107e8","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-21 05:31:25.000000000","message":"Patch Set 8:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/658 : SUCCESS in 0h 51m 55s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":8},{"id":"7be5e01600302e6cb6a62a94d2ec2dcd7fc472e9","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-21 05:31:27.000000000","message":"Patch Set 8:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/566 : SUCCESS in 0h 51m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":8},{"id":"0d948c401252f8e038cad750b064b058a4df7a1c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-21 05:31:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b5569784bdba41d980e3220a66619c46 : SUCCESS in 46m 31s\n- cinder-storpool-tempest-iscsi-multipath build/d0c26abcc2c94fa087ae3f14fa48d40b : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5e0476cc74534846cc997b2af917f0b7c44f3337","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-21 05:32:34.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":8},{"id":"34a32c3facf753199c03213d08c837bab3234df7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-21 05:33:16.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/62fed4b8bfea4ab28c6e6411d3fc02d9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6d0f073f73b4e778fbcc7c2664c55c4 : SUCCESS in 43m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/466f02a624454fae8efe6f2fd1fa6578 : SUCCESS in 35m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e05bb29085d48719133089d6ef4e19c : SUCCESS in 53m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e61ff74c9be4a2ea0793032d6b6aa00 : SUCCESS in 34m 41s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d11cb90facc64310bdd0ae73ac468b0f : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"28b55ae1a33035ef72c13156b3eee0668ccb925e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 06:07:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da38a4ea33749baaf871f451486ac67\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b00b58f621214842b7f582dea7815195 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c2e076b51c54936aa6b8a65b4088ab3 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1d096ed8852b4617b50dbb7fdd065b64 : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"073edb8471282342abf4ad15771cffb7b35ffc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 07:24:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d158db758858405b93a58720378c1dfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2935f1b28f1d4036b5d049c59e4dbe1a : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8954bb25ee344f18cc23d87f6bee2ae : SUCCESS in 7m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23713921f1e3419bbbdf1dab6beb4755 : TIMED_OUT in 40m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d034dc0ed1c7454fa660e00461a245d0 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab62fab3b3340ac94ec7e05f6057e3e : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/a0ba7d1d2a284b8bb3a0b9d594952fe1 : SUCCESS in 1h 05m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3a0332164ce47ad8b68183ff120a18c : SUCCESS in 1h 08m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f910576d5ac94dcbb18f1728dd568edd : SUCCESS in 1h 22m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc9590ee19e64d818c46c5bed0f1be19 : SUCCESS in 59m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6455c1388814880b955ca4ba22b5448 : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4ef5aa8ab0a4f74824146ac191d1ca8 : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/372b7231c3564119bac69db4aaafb6b7 : SUCCESS in 2m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/082c38ba619a43f389829cdb2d25b7bf : SUCCESS in 3m 24s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/1ca3dd4ee00d41958dfa16d90d7a794c : FAILURE in 14m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2d97b9eb52845e9b820cd19481e6b1f : SUCCESS in 5m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0c7ad59397e947ca995eca7b4ec8cd1a : SUCCESS in 5m 57s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d2ae33010d574e17bcb1b679ef4afa48 : SUCCESS in 9m 25s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c2c6174f7d5c471bb7046eb5c19204fa : FAILURE in 51m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb6422ec53f34e47bd1a05e9f84354fa : SUCCESS in 3m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8a494cf853a4370b2abca46e08da7bc : SUCCESS in 54m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/64912f5e656146f8b208ad0db8d0a59f : FAILURE in 2h 13m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d2a13a0ad5b441a9bca430bc3302fb8 : SUCCESS in 1h 31m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/018001c88ace42f0a8f5088dfa4f8398 : FAILURE in 2h 39m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/92d07c120e824dc9bfe3a3f4919bd672 : SUCCESS in 1h 25m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ce67cc528ac04dd185be1382f13b2e95 : SUCCESS in 30m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2274d5c27d234f61833b8656a6d1156e : SUCCESS in 42m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9066cfded444c078b3b418b46220fcf : SUCCESS in 1h 37m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8c2013141b044225a872619d94ee090c : SUCCESS in 28m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/89cb50baacdc4462bc2dc695ce51db6f : FAILURE in 1h 50m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2349d299b094cd88891cbc32f4b040a : SUCCESS in 46m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb71d3bb2bbe4a7b9620f5e06aae0378 : SUCCESS in 29m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"e489c68b507919e6ab51534dae18a525083774f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-21 10:26:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"ba1c46cf72177cff74e2da68885ceaf509e6806d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 10:38:57.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4686/.","accounts_in_message":[],"_revision_number":9},{"id":"2eaa57095a82a4393b8d1e92e43ef086bdc0dd5a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 10:42:31.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4862/.","accounts_in_message":[],"_revision_number":9},{"id":"d2195976969e8e8856ae3eebaa4394010e427e28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-21 10:58:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/988315 : FAILURE in 30m 23s","accounts_in_message":[],"_revision_number":9},{"id":"e9decb293573c909698933d760160744ba22cfce","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-21 11:13:30.000000000","message":"Patch Set 9:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/568 : SUCCESS in 0h 46m 38s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":9},{"id":"4e49cc619c22d058fd3f34d4d275964046bee282","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-21 11:14:20.000000000","message":"Patch Set 9:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/660 : SUCCESS in 0h 47m 32s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":9},{"id":"c314433eb419f486c62fd245cf4b2610fdf0e092","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-21 11:15:18.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/923 : SUCCESS in 0h 48m 25s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":9},{"id":"429ca95763113b43e740a8657bd46912b54a0c99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-21 11:16:55.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ea17c2ad1aa14ed2a70ca426479e798c : SUCCESS in 46m 27s\n- cinder-storpool-tempest-iscsi-multipath build/1ac38e056b5e40859755bc3e0c09a473 : SUCCESS in 49m 15s","accounts_in_message":[],"_revision_number":9},{"id":"4dc5b41529034c89d309ac1c41788c896ddc1357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 11:18:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ceff306c3904f3186a3fe570c8fadbc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18362f8ee90448c7b04291f2034eeb4d : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0f36feae83849e7857bb25ec750fc3c : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec2af52127dc456588dd80dcc0709c0a : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0cb9cb1037b44ac13f450e4b18fa3443b1e3fcad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-21 11:19:54.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4ef64a8ae96d490884229fe5b80d52ad\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b636bac426194641ae36eac6d3d209c4 : SUCCESS in 42m 58s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fc7b381ec664d699ac54a1415604026 : SUCCESS in 36m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86b1797fd0574e5f9e343f02ee88aecc : SUCCESS in 47m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f018c4c3e324dd585729566307534eb : SUCCESS in 37m 25s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1148a5b1ad2e433f840aa65d3d363509 : FAILURE in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f11b6dca8fa040d951e6600a909edaed71e70049","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-21 11:20:02.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":9},{"id":"3d8e351e51fc26be011c7eb72b369f669790c350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:37:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e2597677c5439b86c8e1c6fbd75ab3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee3dd1b8ea742a7bbe67507a23b2ea8 : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04789a3faf8b4db5b3e42160b5834e53 : SUCCESS in 10m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aca954672f4842bdb5a57dafc5eba5da : SUCCESS in 10m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2be602477af644e29ae848c6549f7fb4 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f19eeb5ff9c419191855133d72b3d3d : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/084a99b2d2ce4a839e05f9e4ddd7b93a : SUCCESS in 1h 05m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f120d33fa934671af96c1837256d5b8 : SUCCESS in 1h 08m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b235ac7428e4f19b430a93b68eee28e : SUCCESS in 1h 32m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59d7d829e0b6473f8b0aa1627775a86f : SUCCESS in 1h 00m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df459c9d58284b54a78f5622c4b5f3fc : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/876f1ec8c4c0442e8636a27bb582b241 : TIMED_OUT in 41m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/090d99bc40394d55995b2c91cd6b4ac2 : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb999dcba14a4a8d82566514351e84cf : SUCCESS in 6m 30s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/1de94b1352e149a0a03d89abbb1b7a70 : FAILURE in 16m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ad0bf1fd7a545399589727057a335d8 : SUCCESS in 8m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c9115c5bf9e49549e421ef01e9e9aed : SUCCESS in 6m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/538539780cc0402fb862b5dad994c606 : SUCCESS in 9m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c23e657a3b9c437c9a0a6e5e8f4812a6 : FAILURE in 50m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8465fc6d410b431d90d69e187478df10 : SUCCESS in 6m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2e1e125e3a546cfa33c23b530c75497 : SUCCESS in 1h 35m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65b29d045048402e88f2104e024f2603 : FAILURE in 2h 05m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4ee35621bd7b4f81a505adadf8f5412e : SUCCESS in 1h 29m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f5c67907770f444d8b2ef98d7aeb0f09 : FAILURE in 3h 04m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/735802e715fa4ffe84386c76137e8208 : SUCCESS in 2h 03m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/37af45397952466cad0e2c642d016b49 : SUCCESS in 25m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d59d1e49e6a14dad991582a5d0e0e92d : SUCCESS in 1h 03m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb5b39fde75d4316b95306523a47f3f1 : SUCCESS in 1h 41m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3941863e739844b6bebb597abf9e6246 : SUCCESS in 47m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa6a50445cf74c7d8925d27bd037444f : SUCCESS in 1h 47m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67c4265284f247bb8570b1c74a8b655a : SUCCESS in 1h 59m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4893015a97ab4486884ddf9abbd89496 : SUCCESS in 1h 00m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"2a413ac62a6ef872e5fb91c7ef0a83c02153b5bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-21 13:50:41.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_07_2026_13_43_58\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"55a2d6c2e53f294fade08d4eefd23ee85081bdc8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-21 20:34:52.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-8/FC/ : FAILURE in 15h 55m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-988315-8/iSCSI/ : SUCCESS in 2h 35m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"b1c08e0baaaf59cb00224635c2cc742b175d7580","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-22 00:32:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-9-fc : FAILURE in 10m 14s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-9-iscsi : SUCCESS in 1h 18m 3s","accounts_in_message":[],"_revision_number":9},{"id":"f927fa21298f12496aa0766d7936fd945d706ebc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-22 17:11:27.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-8-fc : FAILURE in 12m 32s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-8-iscsi : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":8},{"id":"96f5754a98a924816a9dd766ed561d7f7c09e7b0","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-27 14:16:08.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d242b376d4150b6495aad83d762171ab852aae2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-28 04:15:43.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"ea8b473ad41119cab6a597ed76fd17d4fa4f8d2c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-28 04:18:39.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":10},{"id":"7e35a311d7817975c03804081774731c63dd03de","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-28 04:24:52.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4751/.","accounts_in_message":[],"_revision_number":10},{"id":"62582b14b3d615c7798c981e28d49d1c715a6c25","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-28 04:26:09.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4927/.","accounts_in_message":[],"_revision_number":10},{"id":"9cbd4062e103daf9068da2bbcc98b0cf3c922bc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 04:39:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fd736e8bc24569a2e3ac8501fa0e7f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d71ae6d4bf644eee99bd6855a066769d : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78dfd757eb3c4f9f99b88eda4c452f7e : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7699e38806ec4614ae92d041a786eb15 : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4256ca7b29d71e50c099a045a9a296db00afe67d","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-28 05:04:44.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/630 : FAILURE in 0h 48m 36s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":10},{"id":"4b73cba17803cb951b4dfc19045c7f3e042c1f0f","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-28 05:05:24.000000000","message":"Patch Set 10:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/722 : FAILURE in 0h 49m 20s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":10},{"id":"ad9d64fc36b93e7c6822d3abc401efc846706a20","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-28 05:14:35.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/988 : FAILURE in 0h 58m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":10},{"id":"5aa6e1943db2200a348517890b17302b0ef86b33","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 05:28:21.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5ee5e57fb61940c29051b11e96715eef/","accounts_in_message":[],"_revision_number":10},{"id":"de41a4d37569051f40d1d3bebb485eca02b3537d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-28 05:45:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/1066d5dea86848d48e6d9164d5ef862a : SUCCESS in 1h 28m 45s\n- cinder-storpool-tempest-iscsi-multipath build/deac9cc2b7044644afcaaa3f2917ee44 : RETRY_LIMIT in 34m 26s","accounts_in_message":[],"_revision_number":10},{"id":"c828a8c44d7d5892ba46c8c41842bf99d7e808c6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 05:49:36.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5ee5e57fb61940c29051b11e96715eef\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c346f6ad6ad74f9f91d9c857209cc442 : SUCCESS in 1h 25m 59s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95b82b31daf24af8abaf4bee6a7bf344 : SUCCESS in 1h 12m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f29b87c7f5bd44459f4327c2308c0e89 : SUCCESS in 1h 18m 16s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/afdc1f0597984ce490768df687c85265 : SUCCESS in 1h 12m 03s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3bc783029b10442dba8602dfce3533b8 : NODE_FAILURE Node request 200-0000553358 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d0edf7b3dedaaf694f942817b0b61ffab9dfa23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 07:54:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3967b6f292bd418aa4220aacca361a01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c03d328baa44aeb7bdff81f3f23985 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/084245e7b31d41dda0bbbd087b44b0bb : SUCCESS in 9m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d3a8b3154c84dd7bb64f0abd0644ad1 : SUCCESS in 10m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/098c28d4202e40a782e9c7486d16c5aa : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4717b9260fcf45bd8d5b5c17e7ac603a : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/6d7af7dd678c42c4ab105be5f391fc15 : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe417d5311334eb0a257a38ed34082b4 : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b373ea8450b14bbbb4d0313c352fafc4 : SUCCESS in 1h 02m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6b7a57043dd4323bbaa40a888943338 : SUCCESS in 27m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e387bc3bc61745b298137eb0b1431c38 : SUCCESS in 5m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28a4aa4cccae4092ab37e9b84f00a25a : SUCCESS in 19m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/401652a3b7c147b1b0a26458b025fb32 : SUCCESS in 4m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fbb54559507e486ebb7440cd64fa3b2a : SUCCESS in 5m 04s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a21ad346c21449bdaf0e4a81b9f50042 : FAILURE in 8m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ef49bd1e44be4a5daa31b9fca63eeb8d : SUCCESS in 5m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e3d93618cd174e649279e98c8ac82c47 : SUCCESS in 4m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e6e1146ef7544389bcb453f24fdbbecc : SUCCESS in 7m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7eb8d30ef534c9991c095222a1b95e5 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2f517860d7b344eba639ea9991b6159e : SUCCESS in 3m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86cf6fdd232a4763be84c92f6d0d8602 : SUCCESS in 1h 38m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b4f177202966486081814a031e652be6 : TIMED_OUT Host unreachable in 3h 33m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dfd4828e8aa14ba38b9d23eaf8769695 : SUCCESS in 1h 37m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/4649e4f8716b45fcb8dc7ac123b4a6ad : FAILURE in 3h 23m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3d4f935dd7974ac4abcf9da3bfe73e32 : FAILURE in 1h 55m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a74d73d5dd67404698461317d4b75454 : SUCCESS in 19m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/816ce67dafee4e349c77132baa6b8b37 : SUCCESS in 37m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/baf4d17e266e40588b753eb788913bd9 : SUCCESS in 1h 15m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/413742d309ce48f4a0ca608406fbd0e3 : SUCCESS in 25m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0e04bab828144be2bc8c9c745493e121 : SUCCESS in 50m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0c9213e53ac4d91a090e52ca06d0fa7 : SUCCESS in 44m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89a59c019b85407b9c803c5d87613aa9 : SUCCESS in 56m 04s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"0ae3fa215de3bcca7431f9e352bb54f65460a364","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-28 08:09:25.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_07_2026_08_02_23\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"f1dc54c3c275d5d0fa7d28c7aca22e6f76d990e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-03 09:09:13.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"0a69b3528bafa1e5e82866caa65ebafa854ff998","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 09:22:09.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4824/.","accounts_in_message":[],"_revision_number":11},{"id":"9d47aa581c9b06e35bb7cb022b14e8d66c4ee2c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 09:23:23.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5000/.","accounts_in_message":[],"_revision_number":11},{"id":"56aaa728db4c0c882fec49ed7685712c1f299164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 09:34:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176473f1249b44ba9b1b9f4971fa6e52\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/358c1e3c26ae4f85b4c86f07093e8fed : SUCCESS in 16m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e99bc6897bf3445792b2098bc2d1dc5d : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4fdc2a39f8c445cd87c796337c101fc7 : SUCCESS in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"40b32a67dde8b1e63dc0437d1341231a5c95ce69","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 09:46:45.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8b161e03847640c1bd205455a03232c3/","accounts_in_message":[],"_revision_number":11},{"id":"cef4cd6d705497b2479c8a5fd353ccd5073bc9d8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 10:10:16.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8b161e03847640c1bd205455a03232c3\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d0d95051c4f42cebbbb45d295ae04d9 : SUCCESS in 44m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9fc577f4794141e99492b0acb765172f : SUCCESS in 37m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/900a6202d5bc4adb8b02420527b51a9b : SUCCESS in 47m 14s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/351394a7567d40feb6e538f5ea25b507 : SUCCESS in 35m 45s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f4c943351aae4a22a1710aa799161477 : FAILURE in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ad65061d1729abc4e0be46621e1af12032bc50a2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 10:12:34.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":11},{"id":"65c93073c69637d39da0474cda90ee74d87106d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-03 10:40:15.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/15b9940f82fe47dcae288d7252f1b751 : SUCCESS in 45m 25s\n- cinder-storpool-tempest-iscsi-multipath build/19bcb04b920d4cab841001fe48840905 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":11},{"id":"76f8234ab6f98e6590c13db6370b4d444534a40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 12:32:19.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17c28ccf3ce44689ac1d6018ca7be4ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2d6799805c4ddc96b2032a69cc5eb5 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/154177aff5c8402ab1aa376af4bd3241 : SUCCESS in 12m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20290c02d89a4b58a347cd5ad2d96bdf : SUCCESS in 9m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2543f57477445caad728ed15c992832 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3445f649236d4219b1eb610dbf1b0690 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/7737b2adf42f4bb0937b6a0eef9576e4 : SUCCESS in 35m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/229e87b1a496434e98b5e044eca5fd86 : SUCCESS in 41m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac999ba7145e4bf5a14a0fff6a135f83 : SUCCESS in 1h 01m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1fcf7886141439ca8ea307d7250f9bf : SUCCESS in 27m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33fcb59b1331494fa9ec1411f1744041 : SUCCESS in 8m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc30a17077154ec9a9773954f55cb06a : SUCCESS in 22m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a17e2b9605f84a3fa8a77bfcde9cf841 : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc3b111702644e43a1bef1c68171543e : SUCCESS in 4m 54s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/32a3f2111daf4b1789622b4de30fb4b7 : FAILURE in 9m 43s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ca268f621b34b46ace0a596b05c58de : SUCCESS in 9m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d28bc3a086dc4ed5a72237b566dc24d8 : SUCCESS in 4m 29s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cc5d47315d8347eaa38d0d6c6ae6f411 : SUCCESS in 7m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/312bd07b69e44446ab2f0de1930849dd : SUCCESS in 39m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e215a7bc746f4a4db1aaf7c26bca909a : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3a9152c736384b6a864ec39f82b5e208 : SUCCESS in 44m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3b64ca06d69445c9a20e46663013cd93 : FAILURE in 2h 04m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ae0cdee701b47f8a3c1ac51f47ea663 : SUCCESS in 2h 01m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7a10f5ed52624d9a851982fe701acff8 : FAILURE in 3h 20m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/ec5a1c56840e49868ac4b6a32d160087 : FAILURE in 1h 26m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/89e31f13bb8c4b5c87cfae710e4f5420 : SUCCESS in 17m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bbba825943ed436e844947e9cd650070 : SUCCESS in 1h 15m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f2aca3d926974dffa756db19398df8bc : SUCCESS in 1h 51m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/45eeff0868cc4616b4257575e6834eeb : SUCCESS in 45m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aeacf872c83d4858a06331019e63839f : SUCCESS in 1h 55m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0307529e1b2641679509a5fa06a48bac : FAILURE in 1h 31m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a5db2402ce9464f93f55b0a68d82188 : SUCCESS in 39m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"a2650848d43d63efd643125a93fd571418f30fb0","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-07 03:27:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"82167615aee6c908bf871f48db7b12059293d0a5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-07 04:32:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/65c6c89874b84c65b37f0d290cefa58b : SUCCESS in 56m 01s\n- cinder-storpool-tempest-iscsi-multipath build/fefcb589aace4799811615092635394c : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":11},{"id":"aff7e64b60ed26549a5a32025eee98a87af04c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 04:49:45.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3452f12b6994491787c303476ffc52ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b80090e3afa846349004e66db5e97ef7 : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/463cf3b5d6fa4539aa5e3f531a88c0f0 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/02b40c187a984f56a0a4e1b92fed802c : SUCCESS in 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cfbb665460f75742a33b9ea23667d73d88076fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:09:56.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57a882463dc34b969462ad07af36d4be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cdd305b2244374a59a30cb64e9075f : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3905340d34f441f8ab4fed47c9c20d98 : SUCCESS in 10m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/183865aae622458f80ee1db936fb25ac : SUCCESS in 10m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/962027be561e4c65853b1253e9cfd0f6 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b4ca22090b48779599d9882ee18d46 : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/996a4768ee44443c811d90e520a037ec : SUCCESS in 1h 10m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/346c67e4b43a4f969a6e945be60ac716 : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f25a9f281ca4ecb878a6fb7b72055ab : SUCCESS in 1h 41m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecaebfce24e74c7d9942f5fbc7fcb19b : FAILURE in 1h 01m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9053d083a9f4be3998a88d012209a18 : SUCCESS in 9m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f61c116a0044df89f1073ad01298822 : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fa8e0e734e034c0ba4ef7edcc314696e : SUCCESS in 2m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6ae04c3a7fd43ca82a67d9032781ac6 : SUCCESS in 7m 56s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/c148ca1901af498f914a7774f6d398b6 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/79650c1251384a6490ed1708ba84ae9a : SUCCESS in 7m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7df6ba563c774f5eb21f8821d07f8b9f : SUCCESS in 7m 41s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ee5a7b5e676841f580ab0bd154cb5d8c : SUCCESS in 10m 40s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/35da96ad45c241bea55b430ccdf61b39 : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae723b916f5547ec96406a9bf1d1ffaf : SUCCESS in 7m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2be83997efbf41be87d856bd8c3f0fb9 : SUCCESS in 1h 32m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d275f344af2e415b86b08dcb654de3bb : FAILURE in 2h 00m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5f0ee68d82144cc4906aefce5a67a470 : SUCCESS in 1h 49m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5a8d52dbe9e748a5a4cd280b0ddda4df : FAILURE in 2h 41m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8f5128eb3d204d46a2a86adcc5d96276 : FAILURE in 1h 25m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4630df333fbc4e48ba7f4b7835b4b9bd : SUCCESS in 38m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16519312c11b45a092349067b073a498 : SUCCESS in 1h 14m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dffd0ee645944d0ca7c00dd35ce22ba4 : SUCCESS in 1h 28m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1ee434cebf4a4b639ccbed66d7209bf9 : SUCCESS in 48m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3bd094fa016e4ed79d11bd994bc496dc : SUCCESS in 1h 48m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9f8d0145db346549c682ffdace56fbe : SUCCESS in 1h 35m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30d13902e55a4d88bd699fb8dc8c67a8 : SUCCESS in 26m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"ea3c3cf45b7747cff698ebdca807bb81f20f4542","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-07 09:22:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e86174b4c939648bffae5defce5351a2b527c96d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-07 09:45:32.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f11f5359266447a6b7a1654a91beba89/","accounts_in_message":[],"_revision_number":11},{"id":"bcf821dae65832c280da7c2cf0f1069cfd7d2b47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 09:55:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1229c014e204b2d9832d010633eab3e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c958837f3fed4849a4f07624c58f624d : SUCCESS in 23m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1b24b40af8184647bf6305acc214bcc0 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ed6e742f93494c028e4eac984de66836 : SUCCESS in 29m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aa49be0e60097432e081bc39aa47ef5df44b0aec","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-07 10:16:04.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f11f5359266447a6b7a1654a91beba89\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ad592983c6849b1a7d6698c6cef1cdb : SUCCESS in 36m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d5f019c364143ff82c2fefa5b363bd9 : SUCCESS in 41m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bfb31a068c374212bffb017ad249afc4 : SUCCESS in 53m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb20861ce9494704a3eecb89838427cd : SUCCESS in 35m 29s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca3551894ca54303bd4ef82555fa9893 : FAILURE in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"16e38b7fde4964dbd5ad4d89f5428edbb872a34f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-07 10:17:28.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/19b715f5e1104a25bcf60981fdd331fd : SUCCESS in 50m 22s\n- cinder-storpool-tempest-iscsi-multipath build/67ac5321270f4c0ea0cc4012d50cfe40 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":11},{"id":"a1d824ecb3b6a7c27f2c5c8a7f0193aec2118c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:12:13.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ff4288c3b554036b1bfa52d8af32667\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a8f582ac304eadba3ba51ee90ae968 : SUCCESS in 8m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d42aac2b75942f5ab8eca2e9562d2ff : SUCCESS in 11m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38848b329f9249b38cd69921636a803b : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a00cd80bd2384ded873592e786f14a8f : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c412305a02f74c6fabf2c45afb12996b : SUCCESS in 14m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/5756344157954cc89ebf77b7179d5118 : SUCCESS in 1h 01m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de86da79f66e4eec8e95489868ae181b : SUCCESS in 1h 05m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1cb5556d5017489a89667b985c126168 : SUCCESS in 1h 24m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eec8bd13ed9f45bf9279aefe2e54c5d7 : SUCCESS in 54m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a0c9a8ae844e769a98fb7ce041cc1f : SUCCESS in 9m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6b0b8f9236d4afab1544d613cb6b4ec : SUCCESS in 23m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23444a51f2274134b46b96ec20a99bab : SUCCESS in 9m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/870c48643d194db0a06bb36a32a58a6c : SUCCESS in 6m 11s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/9bcfe306d64a4097816f91026466d138 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3fb088fde1e645069b274d090a0ce014 : SUCCESS in 8m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5475f0814150474e8e8dca364324f628 : SUCCESS in 6m 26s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8ea8edd18ca349a392c935be4ef75b13 : SUCCESS in 9m 25s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/34cdecdab6f942f4899d178be87efe47 : SUCCESS in 30m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e13d630c75a84247a74db834a5da4c69 : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f39e4ed82a94c04bc2d6fb643cbaebe : SUCCESS in 1h 28m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/781abb11b5fe49c99a90724702047e9a : FAILURE in 2h 01m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/540d2fcba9dd46bcb2c4bff64b897f35 : SUCCESS in 1h 54m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d18b7f63d67e4a16876fa511937a3046 : FAILURE in 2h 38m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/28894e8875ef465388eee780683e3187 : FAILURE in 1h 50m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c46f6164b09947a7bf6ed026b2984be1 : SUCCESS in 37m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/393919642ab74761ad1b9c2ade8ce549 : SUCCESS in 58m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df31c1d722744590bb8e138b57bdd914 : SUCCESS in 1h 32m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e17339c039e4c86b18c925babf8e40a : SUCCESS in 44m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68bf387bd1c648a28221b02753531400 : FAILURE in 1h 59m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2058694f2bac4b3ebd644f5db254ab26 : FAILURE in 1h 44m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/596551e851704334aa7ee4b2479b9d4e : SUCCESS in 52m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"bd19f64f1bb828508db2c9b5bf3ad23c79c0dcdc","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-07 16:02:03.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2e88633680f9b083eaf908635f0164a98c3392fc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-08-08 07:21:33.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/988315/11/20260808063507/NEC-ISCSI : FAILURE in 9h 02m 17s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/15/988315/11/20260808063507/NEC-FC : FAILURE in 9h 46m 24s","accounts_in_message":[],"_revision_number":11},{"id":"730019647a5c9582f89e509433659652aa85b125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"date":"2026-08-12 09:38:47.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"88f4667dbdfa1047ec99849b0be5e008f2c9f380","author":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"date":"2026-08-12 09:40:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"abaf87b2dfe5a51829e1fbe9b3fd8c78db15f0dd","author":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"date":"2026-08-12 09:41:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"44ece240cc6db1332138139070bca8c4c5cb4335","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 10:02:52.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3fdb87031fac477a94e2b9d75c1b6218/","accounts_in_message":[],"_revision_number":12},{"id":"76aee24b1d6769f91bbd059d7292a25e961689f7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-12 10:12:32.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi-cross-pool-cache https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/12/upstream-check-cinder-cross-pool-cache/cinder-tempest-plugin-ontap-iscsi-cross-pool-cache/3389e23/ : SUCCESS in 30m 22s\n\nComment \"Run-NetApp-cross-pool CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"69b73ce1640506114b3ecb9bbc3f3fca363e2e12","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-12 10:13:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/12/upstream-check/cinder-tempest-plugin-ontap-fcp/612ee68/ : FAILURE in 24m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/3c88392/ : SUCCESS in 26m 58s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6cfdb54/ : FAILURE in 26m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"a87ecc4a6792e1647b30ae2fb045794b09db2a3a","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-12 10:28:29.000000000","message":"Patch Set 12:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/885 : SUCCESS in 0h 49m 18s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":12},{"id":"adc512e952764518d7d058946b0a950fcb63d95c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-12 10:29:04.000000000","message":"Patch Set 12:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1152 : SUCCESS in 0h 49m 49s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":12},{"id":"625a761253a29fbb059ffaad63e7db91f1e0de24","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-12 10:29:18.000000000","message":"Patch Set 12:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/791 : SUCCESS in 0h 50m 6s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":12},{"id":"ee6359ae4bc828ed4733e71b0fe5a9f36e49c9c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 10:34:26.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3fdb87031fac477a94e2b9d75c1b6218\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/45c82bbda79247358d10aab8a874c7c3 : SUCCESS in 46m 08s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a72754a7b9c24e3e91b4d40216f52cb6 : SUCCESS in 43m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0140024c917f4964ac4b3711fb869ac8 : SUCCESS in 45m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9f96327196f14dec8bb5eaa0eb8803da : SUCCESS in 42m 05s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22a90d8831f543a7a3ae1a7452468a82 : FAILURE in 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8610f67112f73fa3b4a9d9f7273bc5dd6f87f2aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 10:35:39.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":12},{"id":"3bacd81f6a8942de1461786aed1fb916c7b0c8f1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 10:41:10.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4921/.","accounts_in_message":[],"_revision_number":12},{"id":"5cee1d0439fdf15b1ebe5014eab3af92a2d7424a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 10:41:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5097/.","accounts_in_message":[],"_revision_number":12},{"id":"4b8ecf497375460bcc71a53e884115f40b6fe54a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 10:42:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb8d50d8a9b2420193800de6dc7582a8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30b7d015c1934d279369964f7abcf766 : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d7e1655c362427eaacf8249fecbd73a : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9f18bc1d3b1e4354a3a380954dbf1da1 : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b7f4fe2c8594bf47ae1b1a766a9efbfba338c87a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-12 10:45:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cced582992b8822e3e4c120fbab9527ef5bc2cbe","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-12 10:55:00.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"3193c4e11dc770dee3e2124d789f5d7d71452a16","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 11:05:11.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/83975112aab844bd8348189afdd99e51/","accounts_in_message":[],"_revision_number":13},{"id":"cd2d7cca79e20d7293a9cf709cab01ff68d501f7","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-12 11:06:17.000000000","message":"Patch Set 13:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/792 : FAILURE in 0h 20m 25s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":13},{"id":"ead53081e8a31cb1611ec836e5d61f58dba52f00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 11:13:28.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35a80e84d8b472b86e51f2508544a87\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79338ec6ea2444c486e1a52fc426a49b : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b22738f42f034700b704acc6ce5ee6cb : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf8aedfa0167443a89525f42d95857b3 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"82f4cca5e408f7ad6865593dc4a092ce12dde35c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 11:35:14.000000000","message":"Patch Set 13:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/83975112aab844bd8348189afdd99e51\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0cfbf81d3a6241e2954ea69820895b10 : SUCCESS in 40m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/879705d6bf90481a8c269957de15e58f : SUCCESS in 39m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4159ebed1eeb4dc2976c5b9f76c355ff : SUCCESS in 49m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d158619b067a4038a11d67ad416d5793 : SUCCESS in 39m 30s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a6f1e8cc4648460998a93e8d1d5efa42 : FAILURE in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"40ed75a82bd8cc7cd176d5fa7cc656b1cd40e69f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-12 11:35:50.000000000","message":"Patch Set 13:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1153 : SUCCESS in 0h 49m 58s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":13},{"id":"2a776102e42c0880e6af87e00a592c1ee7ae5ffb","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-12 11:36:06.000000000","message":"Patch Set 13:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/886 : SUCCESS in 0h 50m 17s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":13},{"id":"1bca65d4f44e31679e1673384c9018094d867ef8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 11:43:19.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":13},{"id":"dd96ec6b79ba7903363dd046fe16020e029652cc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 11:44:41.000000000","message":"Patch Set 13:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4922/.","accounts_in_message":[],"_revision_number":13},{"id":"5ab75fe7429ec5faed27e4b4c8eefc19d3e0641e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-12 11:51:29.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d382d4f6aadb4dcb9ec3bd56a9c2785f : SUCCESS in 52m 20s\n- cinder-storpool-tempest-iscsi-multipath build/c5b01c37d9c64af0bfa7327e8be42e7e : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":13},{"id":"f6f8bb2350418026cc518d1a63e563db60c9c4fd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 11:52:47.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5098/.","accounts_in_message":[],"_revision_number":13},{"id":"e238f557426e7aa3528f6e818f2a38ee723817ff","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-12 12:06:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4f4d38b7543cf7a4670878fa6847b8f41721f351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 13:50:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9494316532a641c6b97f85d68c4e663c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e13e1afb0aa4dc3a209733782af2e6a : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b7aee8e32674676bf9c3b608a4af5e5 : SUCCESS in 9m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdb57f4ba72c42b6b18616a7a3cb4bcb : SUCCESS in 9m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20297919133d4257a1f9e85ebfa2f391 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b2ccc4fe944301a663394558771f3e : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/7fb468c1799e4c5ca04abd6435a864b0 : SUCCESS in 30m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/291d4eaabb2a4dd9a8a4c6275c94916c : SUCCESS in 37m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75af9930253740ce9d8452e629c241d7 : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7789db67dc464dd983639784a2952a11 : SUCCESS in 1h 02m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd9c38aebf7a4460b8e05e1349cac41a : SUCCESS in 5m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9be9d98ee02a4b52b3324822d4ce1a1b : SUCCESS in 22m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f334182f634a481f961350eae86649b4 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78cc6e3f62664b16ae46ee61f1f76021 : SUCCESS in 5m 53s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/91536ef9cce7461483a7f1829b7950e3 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6ee0a7e0d5664a23b400aefc2e8b408b : SUCCESS in 9m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cd4c29b06f2d4e918cb2213ffc7d7d2c : SUCCESS in 9m 06s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/30c8964b31a9460886d1b8e65dd3b321 : SUCCESS in 7m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/967689949a414f7f9dba82cd924d2160 : SUCCESS in 28m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/560b3fb08f0748aeb3fbd3c9243598f9 : SUCCESS in 4m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2d87c08ab41246d8929541803624b232 : SUCCESS in 1h 31m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/05da3f9d8b8349a4a3892f8074f2e646 : SUCCESS in 1h 04m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/651ac07d69484387b6963ae7914dc35d : SUCCESS in 1h 27m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/bfa096c833754c62b83917afaba24645 : FAILURE in 2h 59m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9829ede6a5e24d9a9c57ad208ed2ace7 : SUCCESS in 1h 47m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0305520b416849689f052b7488e630f7 : SUCCESS in 38m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/169e0034a1d44b84bf921c75fadf3321 : SUCCESS in 1h 18m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4437fb538d6d4bbda7d7a5962f6fb0ff : SUCCESS in 1h 05m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4a31ae3b402341eba7b76cd41995c5eb : SUCCESS in 31m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7836fab4f97940c99373361c87a69f7b : SUCCESS in 1h 54m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbcd8ca43b6c413d9ef07663751cbcf9 : SUCCESS in 1h 39m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e29a1994dc844679bb171027434bbe26 : SUCCESS in 32m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"46a7e66c4aa04e09228b467ab387b815df683129","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-12 14:01:17.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2026_13_54_21\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"4b37fbcef8cb9f12edf0a005f7cfd1caf2af8ddc","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-12 21:08:11.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-13-fc : FAILURE in 10m 35s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-13-iscsi : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":13},{"id":"fc8cc5e528dc254d6b9efd36c1ec164a1adac971","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 09:25:00.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"b0698e97dc5bdc5f9a4aa578a0a2729ed0853b87","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-21 15:51:43.000000000","message":"Patch Set 13: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":13},{"id":"7b37e79d59935b5ab145174034128623333c7c2e","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 22:57:49.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d1896cd54f7ddebd438c22fdc351459dca4b31c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-25 04:41:06.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"575cc002d2ca379b22d30de687709897c2ddc31e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-25 04:49:56.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5246/.","accounts_in_message":[],"_revision_number":14},{"id":"f562d76d536ce57fa8f8e7b2bcfb5a4bc9c34b99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-25 05:10:20.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"2cf1ce14c44f040bad250ced8ead2e90b1e4d819","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-25 05:19:23.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5247/.","accounts_in_message":[],"_revision_number":15},{"id":"c13cb08c42879f7c93a05ffaeb0f9f7c2ca44598","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 05:36:31.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":14},{"id":"d680803e31734eddeee6e97bed4074533f068829","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 05:42:33.000000000","message":"Patch Set 15:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/90593a4d9a3a462b859fd1063d50eecb/","accounts_in_message":[],"_revision_number":15},{"id":"596771a75c2de2c3b0496ea88dacc5edea450092","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-25 05:57:52.000000000","message":"Patch Set 15:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/933 : SUCCESS in 0h 46m 29s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":15},{"id":"481b950b3d59f254766296d4d697e1c2665efbeb","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-25 05:59:58.000000000","message":"Patch Set 15:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1026 : SUCCESS in 0h 47m 34s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":15},{"id":"603621696054945bf77e7982b47792470224a115","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 05:59:59.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/90593a4d9a3a462b859fd1063d50eecb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c9d6ce43671403f963ca715d90eae87 : SUCCESS in 38m 56s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20646ad36ec04e639058b7b6a0be0b67 : SUCCESS in 31m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23179e14ef9a4015bdece93c96b1033e : SUCCESS in 48m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dbb24b4cca4d498dae50b2d434fde41a : SUCCESS in 34m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/155e4a75e9464b0099cc038645cf9b9a : FAILURE in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f0263d5f223ed1e1cf8815f041354b562e9999fb","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-25 06:00:07.000000000","message":"Patch Set 15:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1295 : SUCCESS in 0h 47m 39s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":15},{"id":"3f345a68363717090f3091670eabc896a1d40222","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 06:04:57.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":15},{"id":"5343b2763bfd19714404291f82eca8d32f0f6f39","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-25 06:59:56.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- cinder-storpool-tempest build/260a4b1ffd384266a2b3dfe5f3a9e9d8 : FAILURE in 17m 12s\n- cinder-storpool-tempest-iscsi-multipath build/6943e7b54d4f42b89f47acc6cc1fe925 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":15},{"id":"db9f22e6fa518d58b5ba60569123d15221d9012d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:36:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b56c744e1647998362abca027521e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d80ab0b9a3a434f99f1a7474412a012 : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c02b805370814a208effe996480d3770 : SUCCESS in 7m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0884a56eca344a23942fc4ee67d26023 : SUCCESS in 10m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff8869a6a6a94d5899df792e8653df89 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba02064bdbe7488ea3138bb8f5040b5e : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/01759b5929d042d18c8856f74bf08568 : SUCCESS in 1h 08m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/968da06b8bad46afa2afb22c8a01fd5d : SUCCESS in 1h 06m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b97ff4c1c60241a88693050b6ba6972b : SUCCESS in 1h 04m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66585981599f46dd8ce10dcf124f90d8 : SUCCESS in 1h 04m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca1020ba6e80418082daa844ccb5a830 : SUCCESS in 5m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6708a0fe0a9e42a29eefe54e0818fe92 : SUCCESS in 19m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3a441ee5112b4faab460d88c69d87b61 : FAILURE in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/865791f82e024f449e22b2c07dfc19d7 : SUCCESS in 3m 16s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/11c9eba386d142cda803f8a68864c544 : SUCCESS in 8m 25s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e14e06335882442abc7dd90d2ec5f2ce : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/386abca7f81c40deacb3a20e46fe0fb2 : SUCCESS in 6m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/09597db34e4948e58fe80c8448fdfe61 : SUCCESS in 5m 29s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7e730fb71ec24ddb9e5754a3cb36f091 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4f2f94b168324db19be2999d60cb167a : SUCCESS in 8m 59s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9e306d6c4e0449ee9b04a76f97fe15c9 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ff278fec8864689a2180a1e63455d51 : SUCCESS in 3m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d225b99dbf204d598212ec6e555f88b5 : SUCCESS in 1h 23m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e73776598e440128dd682405d3422f2 : FAILURE in 1h 51m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de270fe341154fc882c5786e43b8ba5c : SUCCESS in 1h 47m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/619dfb3e1c7c4db0948139267d1d0701 : FAILURE in 2h 20m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e2d129f12518448cbc540a6d100abec6 : SUCCESS in 1h 19m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/40061c7ed60b4cdb9868c4027b888756 : SUCCESS in 26m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/70b9f0c638a44653b5b85cb2e0404bcf : SUCCESS in 41m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/98552b93defd4a539542cc25a8dbbc40 : SUCCESS in 1h 11m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d3a80da77b154b47987d328bdf13a098 : SUCCESS in 48m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a2b97cc95d541ba971b1f24db9326f1 : SUCCESS in 48m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4cef04b987c14df193050e962ef86dd0 : SUCCESS in 1h 23m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4adbb85bf8c34df9a3f2f20a8b1a1d8c : SUCCESS in 45m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"88166041c0e3f184a7ee3ebf7c9782ca2a84080b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 08:40:44.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205482f81897410e98267e49898c8a9e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ddc8844206b434c9d0b0e81c5b9206d : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc1300ad5b154a6d8c3ac685531df4d1 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/affc81b864cf464b986603b4ad5fe435 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6a361e8a88223d60687a01a9d08963ad444d359e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-25 10:59:12.000000000","message":"Patch Set 15:\n\n(17 comments)","accounts_in_message":[],"_revision_number":15},{"id":"81674466e4ba529b4e02e6852547fcad9e74700c","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-25 11:01:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"20b51eb13c1fabec7c15f56f6bd7e64002504f07","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 11:40:27.000000000","message":"Patch Set 15:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/974112810bc64632a94008a4807ca82f/","accounts_in_message":[],"_revision_number":15},{"id":"8096d99228f83142eb549d71fe6d85148c92131b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-25 12:10:54.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/caaa379ca427418f90c97159f0f8bc83 : SUCCESS in 45m 22s\n- cinder-storpool-tempest-iscsi-multipath build/0d31f8d94db44012bb74bcca9d276c24 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":15},{"id":"f40d5ef670f52003e3f4ac7fd8c2ef5ceaed41d6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 13:17:41.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/974112810bc64632a94008a4807ca82f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/83b835eae1114341a940bd55a54b44d3 : SUCCESS in 44m 45s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2d10ea6192884c968e37042a6b98650b : SUCCESS in 35m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b865a2b549e4cfea40314c9866ab4c8 : SUCCESS in 50m 49s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/80a0b6be4b7f466f95c878154f701540 : SUCCESS in 38m 44s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c068240ffcf48ac96d8bf0f2d6a4507 : FAILURE in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a470379670430c1a8383e87e87193b8173c2e2e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:49:49.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af5ec6da9cb84e4ead3728c7e743d014\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedc7af4baad4a2f88d830eb1e6b8fe5 : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f2d7c4419604ed7a85070130a86476b : SUCCESS in 10m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23f2c66509f34d71869a03fe8986a910 : SUCCESS in 11m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bbe19075f9c24171a8c1b0aa6be3717a : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd21ee8a0bd4b24a75e2284d5618cf3 : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/46678045628b47aea6d2dd42f0dce8a1 : SUCCESS in 1h 05m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d300dda48bed4e6d8a29621a39788b27 : SUCCESS in 56m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e47695a759504920af4a2ae0f3b36279 : SUCCESS in 1h 36m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64434fab6b514b4db7d38d44fae54cd7 : SUCCESS in 35m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eac963d6c01e43b3803273955b0adf56 : SUCCESS in 10m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f4c397329fc4994a82d236d7a463147 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e704df3d48734fbfab17fde22cc85608 : FAILURE in 6m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a04e0d976444e30b54ea060410a5f89 : SUCCESS in 6m 36s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d293cbce567e49e2a3d33f44c7c10f92 : SUCCESS in 13m 59s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3683d2b207fb4af095201b846981fa26 : FAILURE in 10m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d8565035d0e14411a6944172f338e0e6 : SUCCESS in 8m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a04601bd4cbc42fd9ebc4de4b176d851 : SUCCESS in 5m 24s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0dcb6b7460a8488db3b298ae1900ade8 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3ae7b50bdac9476a8bb78ba372ef8248 : SUCCESS in 10m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51bc60513a0d4659beb505b3fa0bdcdd : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bef91921687045c69c5697093ba6bdec : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5e2d0550f934e2da9ac0b9f38223103 : SUCCESS in 1h 33m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4119c3fc16984353b97eaa6b478f4925 : FAILURE in 1h 56m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68594004693f414fb51f270531b8276a : SUCCESS in 1h 33m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6e796e319ee6407cbeebe126fe2156e6 : FAILURE in 2h 18m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e0ee566f778a4af7937e5291452c834f : SUCCESS in 1h 49m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d8593f1af36e46d9895155e86af876a4 : SUCCESS in 23m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f40db2568f5c4c9987856946d40aed1a : SUCCESS in 44m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ae2dd7ba556043c5abb3f19117cd015d : SUCCESS in 1h 27m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0927e5145f8344ec80b173b85103c2b3 : SUCCESS in 46m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5b2b13878a944218a7ce18d348cca2d : TIMED_OUT in 3h 35m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd069764563444608e0c3061a678855d : SUCCESS in 1h 40m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b0b7050e3e64938a591df2fe5c40165 : SUCCESS in 44m 22s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"5e31d08bae85885fdf9b7f5f78578b931db3e7f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:17:36.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9280e2944f6c4ec88b8174dd8fe74896\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee92b4325fe94c8ca99d82222f423c9b : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1aed36493c514e29b136f3347c08e806 : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a95f56427e5438d9726b673173436d9 : SUCCESS in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"30c9e81f9dd8a0c1697c2e6b80747de3efaf65bc","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-26 10:35:58.000000000","message":"Patch Set 15: Code-Review-2\n\n(10 comments)","accounts_in_message":[],"_revision_number":15},{"id":"663150b4ed7f66b38c4d932e4bd28654e59b99cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-26 16:56:11.000000000","message":"Uploaded patch set 16.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"36e93287b5e5fff94dfad0431ad56a8ae73c1a66","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-26 17:02:52.000000000","message":"Patch Set 16:\n\n(10 comments)","accounts_in_message":[],"_revision_number":16},{"id":"5ef3aba957b9f37ddb63727355e929ebaaba7ea3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 17:07:13.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5270/.","accounts_in_message":[],"_revision_number":16},{"id":"7e428c9effaa16256f675e5f8d2b86fdab96d98f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 17:16:37.000000000","message":"Patch Set 16:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/97b704671cf34842aca0e9f881a4a123/","accounts_in_message":[],"_revision_number":16},{"id":"c97c5eed3603494618634f8ef1d5100415ed1fe6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 17:53:22.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":16},{"id":"40fc180a58f0b8ac78e713e01c31318a84331b1d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-26 18:16:33.000000000","message":"Patch Set 16:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1318 : SUCCESS in 0h 52m 25s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":16},{"id":"ec43718288d69fc9bd6762c1e63e266889f68876","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-26 18:18:49.000000000","message":"Patch Set 16:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/956 : SUCCESS in 0h 52m 9s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":16},{"id":"bb25ffd738641d1348ccfe66123e47eabab2cf55","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-26 18:20:10.000000000","message":"Patch Set 16:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1049 : SUCCESS in 0h 53m 31s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":16},{"id":"22ea558110f227a6e6857ce97d430ec71f8daedd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-26 18:20:50.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e4e9faf008be44d695afbf9c15af39f0 : SUCCESS in 47m 00s\n- cinder-storpool-tempest-iscsi-multipath build/41b7b6b47af740258064aad7f33a8b3c : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":16},{"id":"e3b071887ce69fa24a4d6eb142dab36d173cf5cf","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-26 18:42:34.000000000","message":"Patch Set 16: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f7f1cde0cc635c5cab91e526887efa10fc6331be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:27:58.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf569f737b34515adbd9162d4e305f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61261c5401a64eff897f9e9a63b9d2f5 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdb573377d0e4028a89e4668e2fa2de3 : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2934102acfc240a392b121b84f3cae35 : SUCCESS in 7m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6556fe256274ebcb06908c8b2626edf : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427ac8d57dd24495b51bf644b7d50bdb : SUCCESS in 15m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/376c84da07b7493eb68033ad1f6a7c1a : SUCCESS in 32m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5dd8bb1606147548c01d6e405719b34 : SUCCESS in 33m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c7e0e09e9ac4d64b7cfb1b0c83c8e06 : SUCCESS in 1h 02m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/765cd25120534bb5af590007aa1f8f5f : SUCCESS in 32m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9191ee5907404a408eb79011e26abe05 : SUCCESS in 7m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0ad3e72e4bec46ffa7c0fb1488f791d6 : SUCCESS in 16m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/306c104ec17141ee916dcd7315db462e : SUCCESS in 3m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c80f576aaacb4dd1905ea624a7da71f9 : SUCCESS in 7m 38s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c746d0a050ac4e8ea45b65064ddfad41 : FAILURE in 12m 45s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/102ed7f1f8b049939775c1b66a39e349 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c6a737f89afe4c258bdf3bdf173db49b : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b8ef0e242d04453480817eac875c53aa : SUCCESS in 6m 29s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c12da2cb37de4e5c965e8a7d9d66c21b : FAILURE in 6m 38s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/674bebaf870f4eba88b589093263dce6 : SUCCESS in 10m 44s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c17b28129da1411187c2f4ccffe31553 : SUCCESS in 19m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5c32c449ac94db48c9ee66f116ca5eb : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49db407c828547a4aa34541f3e18af43 : SUCCESS in 48m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dab4e21ecb1f4325a02206a83fea7a0f : FAILURE in 1h 00m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/39582a75c6d3414daf62db8b2049aacf : SUCCESS in 1h 24m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d15179136a9d4f71a520de7b1b22f5fb : FAILURE in 2h 22m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/950eea59791d4cea9ec66f042390d999 : SUCCESS in 1h 58m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/269d7c24f2154a639262801ec540862d : SUCCESS in 20m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a9ec68b1d3946fea5461c16c47fb389 : SUCCESS in 37m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f58fd86ceacd459f986c21a75be28e25 : SUCCESS in 1h 44m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2ca5c6e1f6224053848d44e2aabe63ff : SUCCESS in 31m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ec098b3d193d4a0088bcfa2e91867cbc : SUCCESS in 59m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4dd01dca81b4d7d85a7251493e1b740 : SUCCESS in 1h 42m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f48c1a34b54d648c339761c09b7a7c : SUCCESS in 1h 07m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"487fdda4de338776e9acdd550f702bac6d4cddc2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 19:28:08.000000000","message":"Patch Set 16:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/97b704671cf34842aca0e9f881a4a123\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/928b6e84397447ceafebd51b9233ab6a : SUCCESS in 38m 16s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/08e35da3d4da4590b35eccb9ccc3d963 : SUCCESS in 40m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a351f7d0efb64fe6a83f914f78b6867b : FAILURE in 2h 31m 24s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/203fef77395640aea610d86b740c4ed8 : SUCCESS in 34m 02s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f65a2ba8d0de4f968a2c9a10f33e63ff : FAILURE in 16m 45s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0a5ea9d5f5288ba21b11da99ec7c8f8a5ba1a662","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 19:38:16.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_08_2026_19_31_21\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"c15ce2b87f11aca54007a3d94d57735d35347034","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 22:11:03.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37787975d4ca4d3cb33b8310adf83cee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21ee3982f473417a8e6fea23bb850615 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ec2a7c16261444b83eceb439b6b3a02 : NODE_FAILURE Node(set) request c0aae329c8734c01bf05ec68a487abee failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d486b546ab244bb6b8bdeab77729cb76 : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f479c543c77ebe8f0b35c2586fbb7578ea51b1fa","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-27 01:07:40.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-16-fc : FAILURE in 13m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-16-iscsi : FAILURE in 1h 39s","accounts_in_message":[],"_revision_number":16},{"id":"56498bf3048e4fd9981567e7e2a80320ce011bc9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 01:26:26.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5070/.","accounts_in_message":[],"_revision_number":14},{"id":"969cdefc0f983daadc075ed7a1c4a6bd9f27951b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 03:28:19.000000000","message":"Patch Set 15:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5071/.","accounts_in_message":[],"_revision_number":15},{"id":"0a2d518296d3ebd46423e1482d7fd4598e191a59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-28 05:08:55.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"006ba835284fb43368ce0c63449fb37cca683667","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 06:01:13.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5285/.","accounts_in_message":[],"_revision_number":17},{"id":"45d7b53c18063a460f6bed2bbe00f579cb707e82","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 06:11:10.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":17},{"id":"0b3794946e234d5e38f35737d5e58d9171a1e8b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 06:57:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a70f144ad05c490a89b9e8690f45e3f1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbc91094d3624d3287734a267561c5e0 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/496fb4e31b694ab69db3347c49deb945 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dbac61023dcc4367ba3f6a97c795b4cb : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2b1ad938b3c4055a838918162fc8a63c2433f51c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 07:07:46.000000000","message":"Patch Set 17:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/374635d3245746be83c747480492d9ae/","accounts_in_message":[],"_revision_number":17},{"id":"3b7e495425d35958b957eb2e1bd4c4e6770288d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:13:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a081aeee1a84a37bf2ab95ae658f861\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdbfbbc7d1d40dd9b0cdc83fcce3079 : SUCCESS in 8m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4f1cdcf5dee471fa9f4f05540a0b95b : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/def30bb31c3646ffa920c4ee73719ace : SUCCESS in 11m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/511b587114284d4f8529b48cb459517a : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780ed214fc6e453e9ae55b467910d7cc : SUCCESS in 16m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/2098d67ce88343a8afadb04f13aaefc8 : SUCCESS in 29m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/255934ef044d4879ab970fa564bdcc64 : SUCCESS in 1h 02m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/024ffb08ba834b318aa560ab99d947d8 : SUCCESS in 1h 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f60c5d397d34c1cb5cf6cfb08d08de6 : SUCCESS in 59m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/766f52a7c96044a9a6fb154c67e7ab40 : SUCCESS in 8m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f382099334164b6fa4b7287622d1ab9e : FAILURE in 26m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/095ea00904b6443caac2e42b3c7c4a53 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5afb9be8c5284f8aa11403eef459b5a9 : SUCCESS in 6m 17s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/cb05cf5b1fb1409c947ffdf488c04ccc : FAILURE in 11m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/143a611cf442471396989c9f01383eb5 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a715db5640a34d46abc0738c1438f189 : SUCCESS in 5m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac7f58dae5a24559ae7bafffcb9ae5fe : SUCCESS in 5m 14s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/371bec30adef4cf28c5637007900758e : FAILURE in 5m 23s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/491f6f625e0741c8bc3631e8c7986b1c : SUCCESS in 7m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/419083f5f3b64adb9d89372c5ae2c83b : SUCCESS in 35m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a7ab43f5d474bd683edecc036793304 : SUCCESS in 4m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b84b6b13e6d346dc927bb87027084948 : SUCCESS in 1h 41m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b691b0c9a96245dc8a81205fd4c48f6f : FAILURE in 1h 01m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ed9b5610ce045cf89d564846aa461b0 : SUCCESS in 1h 58m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9620fee58d9344a5819ebb68095f1789 : FAILURE in 1h 42m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dba70a5f3cd245f490ce95b65a9b2001 : SUCCESS in 1h 46m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9181cdfba292497ab11541dc76d25cf5 : SUCCESS in 19m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/78fc4e77d72846e6a0480c7ec395a75e : SUCCESS in 1h 19m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5e7f813d4e84e8d8d24bbdeb3667bd2 : SUCCESS in 1h 14m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f67f070fcf394d89a3404849d33851a6 : SUCCESS in 46m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5bb0be9d58bb4b8d894e5dfce76cd9fb : SUCCESS in 1h 38m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad509ac00d2a43b6aed24cd5b2330deb : SUCCESS in 1h 38m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/592a9f83ac074d5d95f436181bf0c532 : SUCCESS in 40m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"4e78f087f596de0b24ff42a168493cbcb46db992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-28 07:22:43.000000000","message":"Uploaded patch set 18: Commit message was updated.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"50e61ed698c7c0e089b2afb629f9e23e127dc90d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 07:24:14.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_08_2026_07_17_07\nLeave a comment with","accounts_in_message":[],"_revision_number":17},{"id":"af457397c109138ecc0b24168ccd1c140994ba74","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 07:41:02.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5296/.","accounts_in_message":[],"_revision_number":18},{"id":"5aac7c752f220a5a10a051047b5f94d6388c9865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:50:41.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b8c652960fc45a2b6c129339d146ed3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c268b91f7fb24735b648f792b041a12f : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f86ba397f33d4288b97d8504f070c807 : NODE_FAILURE Node(set) request da3821284c3349c0b88677c3d3be45c5 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ddec40dd3e754c78a67de4f22ff47b4f : SUCCESS in 23m 31s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"438e632815def357f32fe39d475a7d1449492a9d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 08:16:40.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":18},{"id":"b502b8c211730ab382fb8e73b1ea430668194a23","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 08:39:35.000000000","message":"Patch Set 18:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1344 : SUCCESS in 0h 46m 33s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":18},{"id":"9481152b37ac3b4aad9bfc37d683afa7ab946753","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 08:41:01.000000000","message":"Patch Set 18:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1075 : SUCCESS in 0h 46m 56s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":18},{"id":"adaf2bfb509180c0a73a19d22897edf949994ceb","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 08:42:10.000000000","message":"Patch Set 18:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/982 : SUCCESS in 0h 45m 57s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":18},{"id":"f8eba9a66142e8ac0dc5961ae96b34486b65d0a1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 09:07:34.000000000","message":"Patch Set 18:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/74aa5922e48247ddb1e3f4d8ec512ad9/","accounts_in_message":[],"_revision_number":18},{"id":"d05eb73495b6ad9adf7479d4c798f40931ef053b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 09:31:26.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e5ec0b34ec48f2b7122e536fd4a2f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e06b322d689446a885097345d285d97 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aca7380d134442f8cfa34c0831d9bae : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7c6967f38854523910e4cc4deb329cc : SUCCESS in 10m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09f2ba39e7834dd2aeea4583f59596f8 : TIMED_OUT in 40m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a696cfa55c614db0aa1d4df43e7b334c : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/4f6dc4236ae5498e8e3bc6965c255bb0 : SUCCESS in 33m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b0aa76243f04bfaae5de627357100bb : SUCCESS in 33m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a55ffa06f744a10a06e9dc86dda6108 : SUCCESS in 1h 34m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e43790651c6043d3a0a2ef55acf0241d : SUCCESS in 42m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25b471c610624dad9444b1f5521bae9c : SUCCESS in 8m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/da6c217ab40f415ca3359d6e2d7b88c2 : SUCCESS in 17m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c65216efb861497fbb3b535e134427da : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b6a0789407b41e2808ea0f9440e2306 : SUCCESS in 5m 41s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/5b5e04890d6c4b15902b952c93bb9970 : FAILURE in 12m 20s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0e26501c33854185a52d189802551a9e : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2aafb9823b284e519424ae6465c6251f : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2ef232be8e7546d787998e388acc4e5b : SUCCESS in 7m 31s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9a5332a4a277424cb74433e31f087a01 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b9c364ccf789405d86697b29f11e08b0 : SUCCESS in 7m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/036fc3e5d32e46a1b9509772944de380 : SUCCESS in 39m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c515b975920b4dec8660ad366c580ec1 : SUCCESS in 5m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b72a8dae2efc487da2aeef051fc99774 : SUCCESS in 1h 31m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1465961d0e6640caa75e160a8fb397de : SUCCESS in 1h 58m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8bf1049e5484b7eb9b830001fca35f8 : SUCCESS in 1h 41m 33s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a5bd9ee8a0ae4666b003742bcc83a4b4 : FAILURE in 2h 02m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c5754b9ede114dff8dc1f88406112abd : FAILURE in 1h 58m 05s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/59a09a848dd042a6b3d92ebf4bb2332d : SUCCESS in 18m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c180530584c344a0a2ba00dab6c0fcab : SUCCESS in 1h 10m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/925acd6fe956402e9c0921cc55432542 : SUCCESS in 1h 31m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8690b75353414e8e8c21e7f24cf017bb : SUCCESS in 46m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6924b3378f3145c382e358459a2b6271 : SUCCESS in 1h 41m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f9a05335c4124d4ebd5562f1eb7b05b2 : SUCCESS in 1h 45m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1cf4186ec214520ba22a6eea2329a55 : SUCCESS in 34m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"c98ad2850402ee2b0ea5282a78ddbd627716c49b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 09:47:24.000000000","message":"Patch Set 18:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_08_2026_09_40_10\nLeave a comment with","accounts_in_message":[],"_revision_number":18},{"id":"59706a8266dfdfa4090d84af796fb38b9f2f9e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-28 09:59:28.000000000","message":"Uploaded patch set 19: New patch set was added with same tree, parent tree, and commit message as Patch Set 18.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"7380c9c5a8d66781a964579e6c00ea6ea6dc397f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 10:09:31.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5299/.","accounts_in_message":[],"_revision_number":19},{"id":"56ea968f7c733bdfb227342f4db87e3f7fd4b6a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-28 10:20:21.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":20},{"id":"c863b68d85846accddbe0d9b9e5383515a907c05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 10:29:12.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5300/.","accounts_in_message":[],"_revision_number":20},{"id":"eb594f631ff9cd630c630fafbeedfc5dd7d74f37","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 10:50:39.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b118ac51c9a54c5b8d48f98fcff7b702/","accounts_in_message":[],"_revision_number":20},{"id":"86cf4b2b5e0b44efe2634418c7cb6cdb73a54c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 10:54:39.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f94fe5cf8d24a8d9029562b4fad453c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2fdb01cf421d457088124d9041bba344 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ce526853d2449cf978e736681fade87 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c4cb0381e17749a7802ac49099447673 : SUCCESS in 15m 30s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"37d80b2a7a755669657eb287bca098c231e71d77","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 10:56:23.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":19},{"id":"2b18f098db6e2c148d9be1eb95743855896843f6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 11:14:33.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":20},{"id":"d9a219a5b297f87946a6863415968298c6df3087","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 11:33:12.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1079 : SUCCESS in 0h 55m 43s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":20},{"id":"27ae9f190e6c64b0e77c45e3d80928527f1e7191","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 11:33:39.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1348 : SUCCESS in 0h 57m 35s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":20},{"id":"00bd3e8cc63b4650eef5aa3e5d7e72aa0c7883e7","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 11:36:29.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/987 : SUCCESS in 0h 56m 49s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":20},{"id":"05f42df895d6fd9d63888102f413ab70c063d146","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 11:45:51.000000000","message":"Patch Set 20:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b118ac51c9a54c5b8d48f98fcff7b702\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1848f807c14240c1858ec67419a3b546 : SUCCESS in 40m 21s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0956c06366094863871a7c6ece069185 : SUCCESS in 39m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/67845b2099df44f08f4dc34ea1e02e6c : SUCCESS in 52m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5dd965e39909466db9668fbecaafc91f : SUCCESS in 39m 38s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16a838a550184fc68ce96c1e09a90348 : FAILURE in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"4b860ebd519543cacb1721202d14bf448a95a3ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 12:23:41.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0a2f52093804aeca42be3f717410d77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419efd3ef7ce40008f76d437257e51fa : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82653c54c8cc4622be153c615af58fdf : SUCCESS in 8m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fefeb7d88f24594b968aebb6090173f : SUCCESS in 7m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89f7744f1e7d492ea702846d36b1dbfa : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4917645765e04c53ba9ed23785b80d14 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/39389604c6f745f09da2357decf1de29 : SUCCESS in 32m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d21dc3909682416bb14e12e3198a3bcf : SUCCESS in 1h 05m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60e7b6b850334410ba32ddfe82a87460 : SUCCESS in 1h 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b85393d083471489631d6d1d59aee5 : SUCCESS in 28m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34012b2f33aa40a39053845a46189ac6 : SUCCESS in 10m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5b4fb8918a5d4d309e1d20b348395d4a : SUCCESS in 24m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e09bc42aba5041cfa4de5a1fdc2b40d7 : SUCCESS in 3m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ceabe0f92c504a89bba1ec93a90cd852 : SUCCESS in 6m 27s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d86d0989b19b47939bd421707be95164 : SUCCESS in 11m 49s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c7bff608581549338db93d94f8a1b8b7 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fa23c82ecfe45938e4aa41e404ec0e9 : SUCCESS in 9m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6d76fa6b43ab4df7970bef8732b3b1e0 : SUCCESS in 5m 59s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/55fdc43cfbf243d0ad1f86447b827c6f : FAILURE in 5m 23s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1e95cbd0caab4311a9cbb823d031ed92 : SUCCESS in 6m 24s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8790c999f2884f3aa5f40ffbe183ac6b : SUCCESS in 28m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/39e9d9ef6abd40d78bdd635e67f42851 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ff0719802f44a5bbbef1a7ce584569d : SUCCESS in 1h 40m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/10338a958313486bbe8efe698ae2cf0d : SUCCESS in 1h 48m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89fcc6c4aa1b4ac1bca5ab1e1825e980 : SUCCESS in 1h 29m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7f6431c2f2f344b98ffee1e88cbb5ca1 : FAILURE in 1h 53m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/18a6b441e7f04448ac1c4bc9353c0851 : SUCCESS in 1h 22m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4b7489ce809341a2a9e0443d7afaa051 : SUCCESS in 39m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c43626b3bdf8422f95148859e964c828 : SUCCESS in 1h 17m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6025834ce2554e5c9c8090922fb39b64 : SUCCESS in 1h 52m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/eae1c0a4c92340468fc3ddcf54d68760 : SUCCESS in 45m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/88c18bf08a2145019904ea35cd9de958 : SUCCESS in 1h 32m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f93dac6c7c1a4ea1ba65a792a9dfd520 : SUCCESS in 1h 27m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6071317f7f39463c8257d7266fe3d744 : SUCCESS in 53m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"ce005b63e70c40553a5981e354027d0d67a7dbc2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 12:34:35.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_08_2026_12_27_26\nLeave a comment with","accounts_in_message":[],"_revision_number":20},{"id":"dcdbeed3f0d906588590d3d9227836bd219aa747","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 15:50:04.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0353c988840e485a8535307e342542e2 : SUCCESS in 45m 09s\n- cinder-storpool-tempest-iscsi-multipath build/803b033989294f9ba358821f6797d9e8 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":20},{"id":"0550914a1302c033cb004a139437913b239bcf05","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 23:38:17.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-18-fc : FAILURE in 12m 43s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-18-iscsi : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":18},{"id":"c22b1dfca242a1430bd0e44c3670bc2ab0385d76","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 04:55:34.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-17-fc : FAILURE in 12m 39s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-17-iscsi : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":17},{"id":"c8fe40c4ef48dbe761e4c7798e0e65fb199a2ee1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 07:34:50.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-20-fc : FAILURE in 10m 42s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-20-iscsi : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":20},{"id":"54ab17514ca9684d8bde72fb04b4ed1e15f7401e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-30 12:39:12.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"f6a118418f6c6705b955f5bd7c76900cce0e0b79","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-30 12:47:55.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5322/.","accounts_in_message":[],"_revision_number":21},{"id":"c1b7b61fb568ed697bd02d70facf5ea3abe951a0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-30 12:56:07.000000000","message":"Patch Set 21:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/9334d43bdb3d424995a695fbd1973b71/","accounts_in_message":[],"_revision_number":21},{"id":"4dee89518e8a94d8b1cb4b12248ff42ebf11f966","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-30 12:56:41.000000000","message":"Patch Set 21:\n\n(13 comments)","accounts_in_message":[],"_revision_number":21},{"id":"d9718a7cbc0d47169e24e3e372739f98d2526b06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 13:04:42.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239a4f51957a4905a019682f668fc265\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c8304392c4a04aa2aa1a1a34be9a901a : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8226007858814f3ba39f1113c720b141 : SUCCESS in 17m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/464ab4c11cc94b0aa00fbec3e34719d4 : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"0d6a6307df8edd9b66ffbf5557a741b8d45aeddd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-30 13:18:05.000000000","message":"Patch Set 21:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1367 : FAILURE in 0h 38m 30s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":21},{"id":"5eceefc8f2dff1496d229bbac42a9651da508d8f","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-30 13:26:09.000000000","message":"Patch Set 21:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1006 : SUCCESS in 0h 46m 35s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":21},{"id":"859b9b37b49378d92dcb13cf1d23219edb606c20","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-30 13:27:08.000000000","message":"Patch Set 21:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1098 : SUCCESS in 0h 47m 31s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":21},{"id":"883f637db472a46bc67d6b7e95beffe1fe15949b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-30 13:32:01.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ac1acc9caa8d4f2b93de41206c456888 : SUCCESS in 42m 15s\n- cinder-storpool-tempest-iscsi-multipath build/ddc400321e2d4a55a70293af8cd1d495 : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":21},{"id":"56a1eedf6c66e8af3c40031a5092a2c2b023eeb9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-30 13:33:22.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":21},{"id":"b6d874722e9a3a9d011e7c5b4e9c7ab13565e77e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-30 14:19:25.000000000","message":"Patch Set 21:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9334d43bdb3d424995a695fbd1973b71\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88aa7f23818b498fa9e97ab262cf0611 : SUCCESS in 47m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5e5e0bf4241f4578b72809c4e585ed8b : FAILURE in 1h 39m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/10e6e55c826b4832a67e32d26a86a107 : SUCCESS in 42m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2253466b89a84af68cce981a2834cb35 : SUCCESS in 40m 42s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/77281bf468754b4fa8ebc5c5b2dda2ce : FAILURE in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"2b85e1f64d5260d0b874412d5fe4e3ab59da82fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 15:09:33.000000000","message":"Patch Set 21: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3433b05f329a4760a69dcf80197dfee1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca563f83e5945eb90153553ed44b1de : FAILURE in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51e9be1b704f4725aaf34330372a0215 : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/747d14fd6a4f4d19b48fe857141f5f6a : SUCCESS in 8m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2e3eb45c8d346cd9ad789c9414a4a77 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4c3f8d4e96438daf792a06fd0e6df2 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b0a5a1b7ad8c47ffbc74897089d88f48 : SUCCESS in 43m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11c37b1e0446441ba275d618fcf8a34a : SUCCESS in 48m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c12c9ce77c6948c19d889d6b1cfad75d : SUCCESS in 1h 00m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1fdb675059d4cd7bed938217dde74d3 : SUCCESS in 39m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfcc08203b2e4aa89bbf183af1340097 : SUCCESS in 9m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3a1d474b8df46c684ff042782b5c94d : TIMED_OUT in 41m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/23b22f988dae468da041678565cc9f1e : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e714c3012e824a9ea70512ac0c6e965b : SUCCESS in 5m 56s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0cf53617019d41f8bd43c34ea2da9af2 : SUCCESS in 8m 24s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8e0c86cfe6e14952be9e9527ea636cad : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e1fefd2088ba4efabe24ab6beead8dff : SUCCESS in 7m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc581a5456f343df9bd9d465089d3e4c : SUCCESS in 6m 18s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bb842a789c444b61b78e142adba03475 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a301201a919a4f4d94b91e5c15c18c92 : SUCCESS in 9m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b17b44ab030844dfae157face49fb092 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8abd67d4e214d2d87e5832026b7f2e9 : FAILURE in 3m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb928d965dd64166ad10f9206f2c12be : SUCCESS in 44m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7f71ee815cae45e881a256aaafcc1c26 : FAILURE in 57m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/503a4710a05a44a192a795ef6b4f147b : SUCCESS in 1h 44m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b405a5c12ba54b03888bf09ce432c11d : FAILURE in 2h 24m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c9fed491c4934f93a29e8fac382356ab : SUCCESS in 1h 23m 31s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/87788a407abb4e2f8f0ddcb5c3435766 : SUCCESS in 26m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/762d08fb032744a6a4790b4e035bc19c : SUCCESS in 57m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0823723e0ff4425795e6ea794639e7f3 : SUCCESS in 1h 40m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1fe17743dae34e3fa2c687e521623c84 : SUCCESS in 24m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fd49b22f394b4bd99e25f477361bbb21 : SUCCESS in 48m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d40d2bd3b7448009231859bdd76f069 : SUCCESS in 1h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37b605a4b0fe4361accb7d75662fa4bc : SUCCESS in 27m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"4320c5045942668c997484de2f52385c117fff73","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-08-31 08:57:01.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"18d326d9cd8de9ccdbb00b9d6caa60a02a3c9f5b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-31 23:36:26.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5109/.","accounts_in_message":[],"_revision_number":17},{"id":"018f87efcec10054cfe9945b47bdfeb38ab87c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-01 12:01:30.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"46e65ab96b0d382640c9900b4fb40a6c5a41b8cb","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-01 12:07:28.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"654b3ac0344b24858d0b50ea4a94bc53fd3bb928","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 12:10:39.000000000","message":"Patch Set 22:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5354/.","accounts_in_message":[],"_revision_number":22},{"id":"0162539163dc5ea7826e98388a8001174d35313c","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-01 12:14:49.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8c1583b0745b4fa204a46443850a142835edebd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:33:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558fde1ea5124579a42122a47714016d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9be926382b8548a0a90a0cad531d941e : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b6ab9a67d0434e01a5de329c51f27252 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7b74b8f0e11842cea429c4ac5f99cae6 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a5bd9e43bab71b5d7628da0d918120913096d05d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 12:41:32.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/67f0ea8539ca4e42a0a3655d38684f9b/","accounts_in_message":[],"_revision_number":22},{"id":"678747f83e5825201410b3f1527ebeff47650df6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 12:55:37.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4941197a2df741f80c317fb9e270f29f6fb085b2 .","accounts_in_message":[],"_revision_number":22},{"id":"ec9de7ae3a07aca0d28c16606092f0a0c395e2cb","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-09-01 13:43:15.000000000","message":"Patch Set 22:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/1037 : FAILURE in 0h 36m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988315","accounts_in_message":[],"_revision_number":22},{"id":"2d6f2c48c7dc0dc9d87bafaf8ee27afbd60dba36","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-09-01 13:50:09.000000000","message":"Patch Set 22:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1398 : SUCCESS in 0h 45m 38s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988315","accounts_in_message":[],"_revision_number":22},{"id":"e8ae4a7f20dca71f16714633727517fbf038145b","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-09-01 13:58:44.000000000","message":"Patch Set 22:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1129 : SUCCESS in 0h 46m 54s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988315","accounts_in_message":[],"_revision_number":22},{"id":"b1f88a6acc9865cf62e217272885c8016ed23df0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-01 14:11:40.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/87bab836df59493a9e190a48efea333a : SUCCESS in 42m 39s\n- cinder-storpool-tempest-iscsi-multipath build/6a415063cb2943a8b916c8e44ff92aa6 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":22},{"id":"1dea17d4a3b3edd8dd0ccc032d9c0edc5482f3d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:16:28.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b585586f4a04c36b9c04417a88a6234\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee783612ea1a4b58a46d96338db7fdfb : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b9724a39ceb4a5ab40023276e9e7123 : SUCCESS in 8m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12841c0890cd468a91644705b9b0f5a6 : SUCCESS in 7m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34420c97cba2419fb899f1f1c835fd97 : TIMED_OUT in 41m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b863f504232d4630a9b68e76afd77dc3 : SUCCESS in 12m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/4a00164cc4c940d88b519d4e92122772 : SUCCESS in 1h 11m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/930a38bf73a34ed0aa2d3eeef54bb9e6 : SUCCESS in 32m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/348d84f064674a2f83acbacfe8ac5b0e : SUCCESS in 1h 18m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82e89bd9694b4fe6b87e2bebafd5e04e : SUCCESS in 1h 00m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2080d645bedd4ccb9e9cfdeab5d34b8e : SUCCESS in 6m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd45e657ae004ef98291df8f0777fbe3 : SUCCESS in 18m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cadd5bc1465d4ca2a5108b9712a30a1b : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4a774055f324576841b269c2c9d814e : SUCCESS in 5m 24s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/6c2bc71d5dfe42c5a0db3fea05dfe2ad : SUCCESS in 9m 40s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/183eb39ace42473997a924012cf42e25 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a790515734f84d0c9128895bc0c35c03 : SUCCESS in 5m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8a0b659c4211426c93886ce49b853b7c : SUCCESS in 7m 17s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4812d78b7f9240e0b09df76eff03180a : FAILURE in 5m 14s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/fc963cf1296d4c7b9e2867ff0eea4455 : SUCCESS in 7m 36s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7558837b576545ed84c3990f3f5f19c2 : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e31e5eebd4541a9a2f209c1f3faf46a : FAILURE in 3m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a05740b85d4b4501aa31b1bb9ae94495 : SUCCESS in 1h 31m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4f63d692c9c644fb9b55c326601a8813 : FAILURE in 1h 46m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f4c5b95924ce45359aa34e6e6316dedd : SUCCESS in 1h 54m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/705a5a206ad24f5796bc98a81f6674aa : FAILURE in 2h 09m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/99dab630e38748bba2b4a1873f42712f : SUCCESS in 1h 47m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/61e08e2ca6f447229b15dece9d2dacc9 : SUCCESS in 29m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3bc3de8ec7b040a482ede63bcf06356f : SUCCESS in 43m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f1c9c0443e294dd1b9b636b431c7fe9c : SUCCESS in 1h 29m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64f62777aebc45d49f242abc830fc332 : SUCCESS in 39m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/edb9b2f6ac69443c98b0fce55115454c : SUCCESS in 56m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/956d07ed3e6348c7bedee515b9a43df0 : FAILURE in 1h 40m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/690922909e4d496e819b3151510cbad5 : SUCCESS in 56m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"e164c051f110ad60a46a39616bc9e4203a8d550e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 15:29:42.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/67f0ea8539ca4e42a0a3655d38684f9b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1b1bb785d434926b2eed8608c5bd700 : SUCCESS in 42m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd643071ee4e48cfb10fc65fb0c01630 : SUCCESS in 39m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c63b01691ae747a9834f56a12592e800 : SUCCESS in 47m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fd9d6e60263b47ac9c881713efca39fd : SUCCESS in 39m 55s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0be91aba6d9c491091681846358bf531 : RETRY_LIMIT in 24s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"378bf5d0c6c0d91f1abc80e1a90c02e001971292","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 18:48:26.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5120/.","accounts_in_message":[],"_revision_number":18},{"id":"5de85f1bb1e5c0cbc902141c157790e6dc7e874b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 20:24:42.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5123/.","accounts_in_message":[],"_revision_number":19},{"id":"cf2c89cb4a7adf80ff39ae57c830080c44035a46","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 21:03:21.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5124/.","accounts_in_message":[],"_revision_number":20},{"id":"03c2304ffe98d34b7c702764da497c24709a70bc","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-02 03:31:18.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"a850fac8605dc52605cd98d0ae73c066758a5aa8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 03:52:12.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e3b9e6bd9b814b8dbf2adbe38f8799b4/","accounts_in_message":[],"_revision_number":22},{"id":"05dcf93aaaf7068bf9858a319c44a1a2de3f1f24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 04:00:25.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee5b060299146b89a287288733f018b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/942b2ceaccb14d299989a1e86f6f8ac8 : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a16f5cf6ec024c23afbd6077abe61015 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/41a0b2b1517c4aa3b89bfd65d83e24f3 : SUCCESS in 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"4c1c5aa4cb623eca582d55e2e2be4e7981c80400","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 04:21:09.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e3b9e6bd9b814b8dbf2adbe38f8799b4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d6299c7c1844cce98aefdca505fd8d5 : SUCCESS in 41m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a98a96df6f7497193aff82a9eb5d866 : SUCCESS in 35m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/42ae67dfbd624b3695f4908053b38603 : SUCCESS in 49m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b1abd437e194138853aaf3bb8315b20 : SUCCESS in 35m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/543ce761f2e443c597973fb68a01ebef : FAILURE in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"380f1d8897646f3701e8c51241203f5ab2f663b1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 04:29:17.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/95a22706ef9241238845747aa2e94d43 : SUCCESS in 44m 29s\n- cinder-storpool-tempest-iscsi-multipath build/7ea7da787dfe467d8413c2df51187de4 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":22},{"id":"0b56650d12ab5493d26bcfa13deff7c07a8d05ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 05:33:47.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe88d485fea45efb8541aa2284efa06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85deb5848b0f479c821f413f6919a493 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84e555424817406695afe0aa733319bf : SUCCESS in 7m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/256f576e8f904a51a077daf9b478e76c : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50e556ff8fb04797bc64103426dd7d58 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfaf1631c1bf4a04bea3438921cb4021 : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b5023833aca84732b5e6a12fe75e5467 : SUCCESS in 36m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f84b2554a7a74cb8af8e9c59ee00fa84 : SUCCESS in 44m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86fa87e8c9f84f208a259151a14987d8 : SUCCESS in 1h 18m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ff56077ab594a6282b8044f9dd20b0e : SUCCESS in 56m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6421b28501cb4935b80dfb3321114ff7 : SUCCESS in 5m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29ed182599ee4400aac7e161cd12d2fb : SUCCESS in 19m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/34b2160296db4fe391ac27c3ba9dbb4c : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73bf327dfaab499f8fb6f35963196228 : SUCCESS in 6m 19s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/22e9e0d438924c2484bb2a88ce446183 : SUCCESS in 11m 41s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/28b80f395017427096e974ff2297eed2 : FAILURE in 8m 06s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/382ddebf6a2c4ab5b6983dcf1ce01c9b : SUCCESS in 4m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/17ca2d17b40e4898a63bb7e70f6ada80 : SUCCESS in 5m 58s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c3f062e5a5c94f209c92a32e065c2422 : FAILURE in 5m 36s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5e87980403b14323a9df0b031def9fff : SUCCESS in 7m 00s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/006bf16fdc27491a86809e46e9e7238d : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b8d37e3698244e393397d164965a33f : FAILURE in 4m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c3718509c824b05b784033f80ad4afc : SUCCESS in 57m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1e0a2cfbbbda481e8652c225cca9e572 : FAILURE in 1h 10m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f027eb17025b4d238b9653573ba5c8b0 : SUCCESS in 1h 27m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7731ef97de814c9e9b18f408feb32d4d : FAILURE in 1h 58m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df7ceea7debe43488bf48e35a1133812 : SUCCESS in 1h 18m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d14413acde934b1a99d6359657e70b26 : SUCCESS in 28m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b89057e0ec87428dbb53c62f6de2c850 : SUCCESS in 1h 08m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1284aaffe8b04d998aff1a84547cc1d5 : SUCCESS in 1h 37m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9a57544587764254bcbcb2ae407720ed : SUCCESS in 36m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/874caae2c0af48a1be9026e1d9839bdb : SUCCESS in 1h 47m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a12ed4e0f07c466d9de1936df5cfa2f4 : SUCCESS in 1h 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a705fe221e43cda5d370b01eef5b85 : SUCCESS in 47m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"ad59a3c5730f4149dc4c6a489060a5404f9271c1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 05:43:31.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-22-fc : FAILURE in 12m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-22-iscsi : SUCCESS in 1h 22m 1s","accounts_in_message":[],"_revision_number":22},{"id":"062130a73d2572b2fee1ec94d58cc42236f3918d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-02 05:43:52.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_09_2026_05_37_14\nLeave a comment with","accounts_in_message":[],"_revision_number":22},{"id":"0fe76f69915a2bccd29d80aa7f6d8bc7d9fa758a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 13:47:38.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5146/.","accounts_in_message":[],"_revision_number":21},{"id":"50d1a702072cac1fb5f8410a96c5c8b5750141a2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 21:18:34.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-22-fc : SUCCESS in 49m 30s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-15-988315-22-iscsi : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":22},{"id":"41cede035c0d13eb4a520903f8f28f870a3d43fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 20:54:31.000000000","message":"Patch Set 22:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5178/.","accounts_in_message":[],"_revision_number":22},{"id":"8b92b166237bdc295e01710014af8da715f99b67","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-09-04 07:04:56.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"98da6c889bdefe23424fb72f822827b540e5a0e7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-04 11:04:41.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/22/upstream-check/cinder-tempest-plugin-ontap-fcp/63defcc/ : FAILURE in 1h 08m 32s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/22/upstream-check/cinder-tempest-plugin-ontap-iscsi/559507e/ : SUCCESS in 1h 10m 17s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/988315/22/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/04e53a3/ : SUCCESS in 1h 08m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"844c4469326682859f62678ec597e84aa51ed272","revisions":{"aaa78ae150018682080b397cc1ac24b82e207ec1":{"kind":"REWORK","_number":1,"created":"2026-05-12 17:56:10.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/1"}}},"commit":{"parents":[{"commit":"51e260aaaf3d852b045c05f55286d59e270d9ca9","subject":"Merge \"api: Require RequestContext.auth_token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e260aaaf3d852b045c05f55286d59e270d9ca9"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:50:17.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa78ae150018682080b397cc1ac24b82e207ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aaa78ae150018682080b397cc1ac24b82e207ec1"}]},"branch":"refs/heads/master"},"7b3a5c39085ade0e897a977ddfeba5bf11619fca":{"kind":"REWORK","_number":2,"created":"2026-05-13 16:31:58.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/2"}}},"commit":{"parents":[{"commit":"51e260aaaf3d852b045c05f55286d59e270d9ca9","subject":"Merge \"api: Require RequestContext.auth_token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e260aaaf3d852b045c05f55286d59e270d9ca9"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-13 16:27:37.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b3a5c39085ade0e897a977ddfeba5bf11619fca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b3a5c39085ade0e897a977ddfeba5bf11619fca"}]},"branch":"refs/heads/master"},"6855e4eebb4bdf4a0192eaf4797ef7af9b640f14":{"kind":"REWORK","_number":3,"created":"2026-05-13 16:57:01.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/3"}}},"commit":{"parents":[{"commit":"51e260aaaf3d852b045c05f55286d59e270d9ca9","subject":"Merge \"api: Require RequestContext.auth_token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e260aaaf3d852b045c05f55286d59e270d9ca9"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-13 16:56:44.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6855e4eebb4bdf4a0192eaf4797ef7af9b640f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6855e4eebb4bdf4a0192eaf4797ef7af9b640f14"}]},"branch":"refs/heads/master"},"ea60738727d4aa828b1793ca53525cdc4bc754b0":{"kind":"REWORK","_number":4,"created":"2026-05-13 17:24:11.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/4"}}},"commit":{"parents":[{"commit":"51e260aaaf3d852b045c05f55286d59e270d9ca9","subject":"Merge \"api: Require RequestContext.auth_token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e260aaaf3d852b045c05f55286d59e270d9ca9"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-13 17:23:53.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea60738727d4aa828b1793ca53525cdc4bc754b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea60738727d4aa828b1793ca53525cdc4bc754b0"}]},"branch":"refs/heads/master"},"b086e25a7038a2cda86699f571218b73ae0695fe":{"kind":"REWORK","_number":5,"created":"2026-07-08 16:58:09.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/5"}}},"commit":{"parents":[{"commit":"51e260aaaf3d852b045c05f55286d59e270d9ca9","subject":"Merge \"api: Require RequestContext.auth_token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/51e260aaaf3d852b045c05f55286d59e270d9ca9"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-08 16:57:52.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b086e25a7038a2cda86699f571218b73ae0695fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b086e25a7038a2cda86699f571218b73ae0695fe"}]},"branch":"refs/heads/master"},"5abdc756a2f3b781dd7341dd68efcedcebcf7b09":{"kind":"REWORK","_number":6,"created":"2026-07-16 12:56:49.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/6"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-16 12:37:02.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5abdc756a2f3b781dd7341dd68efcedcebcf7b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5abdc756a2f3b781dd7341dd68efcedcebcf7b09"}]},"branch":"refs/heads/master"},"1d07e587b4d249063749c74ae37cfb72903c027b":{"kind":"NO_CHANGE","_number":7,"created":"2026-07-16 12:58:45.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/7"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-16 12:58:30.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d07e587b4d249063749c74ae37cfb72903c027b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d07e587b4d249063749c74ae37cfb72903c027b"}]},"branch":"refs/heads/master"},"5ae403a62cf616a85eacc8a3f76ce1ebc68fa82c":{"kind":"REWORK","_number":8,"created":"2026-07-21 04:39:05.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/8"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-21 04:38:48.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae403a62cf616a85eacc8a3f76ce1ebc68fa82c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ae403a62cf616a85eacc8a3f76ce1ebc68fa82c"}]},"branch":"refs/heads/master"},"94e1f80266d12e5746130f84128bf7a5bf52bd00":{"kind":"REWORK","_number":9,"created":"2026-07-21 10:26:26.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/9"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-21 10:26:11.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94e1f80266d12e5746130f84128bf7a5bf52bd00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94e1f80266d12e5746130f84128bf7a5bf52bd00"}]},"branch":"refs/heads/master"},"78764b4c3ad3b49d9646d865896ae49737d3cf23":{"kind":"REWORK","_number":10,"created":"2026-07-28 04:15:43.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/10"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-07-28 04:14:10.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78764b4c3ad3b49d9646d865896ae49737d3cf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78764b4c3ad3b49d9646d865896ae49737d3cf23"}]},"branch":"refs/heads/master"},"08be9dd02477e10a735230adb7717f0948f8b787":{"kind":"REWORK","_number":11,"created":"2026-08-03 09:09:13.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/11"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-03 09:08:57.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08be9dd02477e10a735230adb7717f0948f8b787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08be9dd02477e10a735230adb7717f0948f8b787"}]},"branch":"refs/heads/master"},"307c95ab680001e7eba8197a8916d42f82931312":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-08-12 09:38:47.000000000","uploader":{"_account_id":39225,"name":"Prudhvi","display_name":"Prudhvi","email":"prudhvi.openstack@gmail.com","username":"Prudhvi"},"ref":"refs/changes/15/988315/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/12"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Prudhvi","email":"prudhvi.openstack@gmail.com","date":"2026-08-12 09:38:47.000000000","tz":0},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled, Cinder\ncan now promote a local cache entry on the destination pool after a\nsuccessful cross-pool clone. Subsequent requests for the same image on\nthat pool are then served from the local cache, avoiding repeated\nexpensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\nimage_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nDepends-on: https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/989987\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/307c95ab680001e7eba8197a8916d42f82931312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/307c95ab680001e7eba8197a8916d42f82931312"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5f7aef0c8855b1a77bac24b120c5939060da7b2d":{"kind":"REWORK","_number":13,"created":"2026-08-12 10:45:26.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/13"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-12 10:45:08.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f7aef0c8855b1a77bac24b120c5939060da7b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f7aef0c8855b1a77bac24b120c5939060da7b2d"}]},"branch":"refs/heads/master"},"7ea32ae7243ab47ebee4b452feac6851d6636063":{"kind":"REWORK","_number":14,"created":"2026-08-25 04:41:06.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/14"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-25 04:40:47.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a backend driver supports cross-pool cloning\n(clone_across_pools\u003dTrue) and the image-volume cache is enabled,\nCinder can now promote a local cache entry on the destination pool\nafter a successful cross-pool clone. Subsequent requests for the same\nimage on that pool are then served from the local cache, avoiding\nrepeated expensive cross-pool copies.\n\nPromotion is controlled by a new per-backend boolean option:\n\n  image_volume_cache_promote_across_pools (default: False)\n\nIt is disabled by default so existing deployments are unaffected on\nupgrade.\n\nChanges:\n- ImageVolumeCache.get_local_entry(): new host-scoped cache lookup\n  used as the idempotency guard inside the promotion path.\n- CreateVolumeFromSpecTask._promote_local_cache_entry(): best-effort\n  helper that clones the just-created volume into the cache and cleans\n  up on failure.\n- CreateVolumeFromSpecTask._create_from_cinder_store_image(): new\n  coordination-locked helper for cinder-store Glance images that checks\n  the cache first, falls back to a direct clone, and then promotes.\n- VolumeManager.init_host(): re-evaluates clone_across_pools after\n  driver initialisation so the cache always sees the correct value.\n- Unit tests for all new code paths.\n- Release note.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea32ae7243ab47ebee4b452feac6851d6636063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea32ae7243ab47ebee4b452feac6851d6636063"}]},"branch":"refs/heads/master"},"5afe7a6eeb9beb523f63b33a8547dbd9c0c6339b":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2026-08-25 05:10:20.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/15"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-25 05:07:38.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False), so existing deployments are unchanged on upgrade.\n\nCreateVolumeFromSpecTask._clone_and_promote_locked serialises\nclone-then-promote (lock image-promote-{image_id}-{volume.host}).\nWaiters re-check ImageVolumeCache.get_entry(...,\nclone_across_pools\u003dFalse) and reuse a local entry.\n_do_create_cache_entry clones the user volume into the cache and\ncleans up on failure. _create_from_cinder_store_image tries the\ncache first, then the locked clone/promote path.\n\nNetApp ONTAP REST clone verification uses the LUN create Location\nheader (/storage/luns/\u003cuuid\u003e) via return_headers / return_location,\nthen GET/PATCH that UUID (get_lun_by_uuid, do_direct_resize) so\nname-filtered collection search cannot miss a LUN that exists.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5afe7a6eeb9beb523f63b33a8547dbd9c0c6339b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5afe7a6eeb9beb523f63b33a8547dbd9c0c6339b"}]},"branch":"refs/heads/master"},"6c6942897623a2d604e4410b40a63c021ffa4f3e":{"kind":"REWORK","_number":16,"created":"2026-08-26 16:56:11.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/16"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-05-12 17:48:36.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-26 16:53:28.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False), so existing deployments are unchanged on upgrade.\n\nIf the destination pool already has a cache row, create-from-image\nclones it as today. If not, the cache row is created either by\ncloning from another pool\u0027s cache volume (when promotion is enabled)\nor by the existing Glance download path. Concurrent creates for the\nsame image on the same host/pool are serialised so only one local\ncache row is seeded. Glance cinder-store image clone is unchanged.\n\nNetApp ONTAP REST clone verification uses the LUN create Location\nheader (/storage/luns/\u003cuuid\u003e) via return_headers / return_location,\nthen GET/PATCH that UUID (get_lun_by_uuid, do_direct_resize) so\nname-filtered collection search cannot miss a LUN that exists.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\nCo-authored-by: Cursor \u003ccursoragent@cursor.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c6942897623a2d604e4410b40a63c021ffa4f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c6942897623a2d604e4410b40a63c021ffa4f3e"}]},"branch":"refs/heads/master"},"893467bed7639165a6bae9d209da2d3395247dc1":{"kind":"REWORK","_number":17,"created":"2026-08-28 05:08:55.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/17"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 05:08:24.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). With the option off, create-from-image keeps the\nexisting {image_id} lock (service-wide for that image) and still\nlooks up cache entries with clone_across_pools, so a remote hit\nreturns ASAP and the clone runs outside the lock.\n\nWhen the option is on, that path uses a distinct lock\nimage-promote-{image_id}-{volume.host} so destination pools do not\nserialise each other. Concurrent creates of the same image on the\nsame host/pool wait, then reuse the local row. Glance cinder-store\nimage clone is unchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nAssisted-By: Cursor\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/893467bed7639165a6bae9d209da2d3395247dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/893467bed7639165a6bae9d209da2d3395247dc1"}]},"branch":"refs/heads/master"},"04d2f1b03b15b12c841379dbf658a925d83684bf":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2026-08-28 07:22:43.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/18"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 07:22:20.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). With the option off, create-from-image keeps the\nexisting {image_id} lock (service-wide for that image) and still\nlooks up cache entries with clone_across_pools, so a remote hit\nreturns ASAP and the clone runs outside the lock.\n\nWhen the option is on, that path uses a distinct lock\nimage-promote-{image_id}-{volume.host} so destination pools do not\nserialise each other. Concurrent creates of the same image on the\nsame host/pool wait, then reuse the local row. Glance cinder-store\nimage clone is unchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04d2f1b03b15b12c841379dbf658a925d83684bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04d2f1b03b15b12c841379dbf658a925d83684bf"}]},"branch":"refs/heads/master"},"3e127ca8a6caf2e4f9a0e55af9dbd7facb6a2127":{"kind":"NO_CHANGE","_number":19,"created":"2026-08-28 09:59:28.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/19"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 09:58:29.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). With the option off, create-from-image keeps the\nexisting {image_id} lock (service-wide for that image) and still\nlooks up cache entries with clone_across_pools, so a remote hit\nreturns ASAP and the clone runs outside the lock.\n\nWhen the option is on, that path uses a distinct lock\nimage-promote-{image_id}-{volume.host} so destination pools do not\nserialise each other. Concurrent creates of the same image on the\nsame host/pool wait, then reuse the local row. Glance cinder-store\nimage clone is unchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e127ca8a6caf2e4f9a0e55af9dbd7facb6a2127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e127ca8a6caf2e4f9a0e55af9dbd7facb6a2127"}]},"branch":"refs/heads/master"},"b164026d80c3eb09f092cd503dd03e4d0c1d2053":{"kind":"REWORK","_number":20,"created":"2026-08-28 10:20:21.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/20"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 10:19:12.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). Local cache rows are not seeded from a remote\ncache volume unless that option is enabled.\nCache-entry preparation now uses the lock\n\nimage-promote-{image_id}-{volume.host} instead of the previous\nservice-wide {image_id} lock. That lock-name change applies even\nwhen the option is left at False. The first lookup in that method\nis host-local (clone_across_pools\u003dFalse). Concurrent creates of\nthe same image on the same host/pool are serialised so only one\nlocal cache row is seeded. Glance cinder-store image clone is\nunchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b164026d80c3eb09f092cd503dd03e4d0c1d2053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b164026d80c3eb09f092cd503dd03e4d0c1d2053"}]},"branch":"refs/heads/master"},"c2893a9e2c1d9433f336ce5f152eb12cb037220f":{"kind":"REWORK","_number":21,"created":"2026-08-30 12:39:12.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/21"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-30 12:38:46.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). Local cache rows are not seeded from a remote\ncache volume unless that option is enabled.\n\nCache-entry preparation uses the lock\nimage-promote-{image_id}-{volume.host} instead of the previous\nservice-wide {image_id} lock. That lock-name change applies even\nwhen the option is left at False.\n\nWhen the option is False, the first lookup still uses\nclone_across_pools so a remote hit returns ASAP and the clone runs\noutside the lock. When it is True, that lookup is host-local so a\nremote row can be cloned, the volume DB row is saved, then a local\ncache entry is seeded. A failed promote clone does not look up the\ncache a second time. Glance cinder-store image clone is unchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2893a9e2c1d9433f336ce5f152eb12cb037220f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c2893a9e2c1d9433f336ce5f152eb12cb037220f"}]},"branch":"refs/heads/master"},"844c4469326682859f62678ec597e84aa51ed272":{"kind":"REWORK","_number":22,"created":"2026-09-01 12:01:30.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/15/988315/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/15/988315/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/15/988315/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/15/988315/22"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-08-28 04:49:29.000000000","tz":-240},"committer":{"name":"Jayaanan Borra","email":"jayaanand.borra@netapp.com","date":"2026-09-01 12:00:49.000000000","tz":-240},"subject":"Implement per-pool image-volume cache promotion","message":"Implement per-pool image-volume cache promotion\n\nWhen a driver advertises clone_across_pools and the image-volume\ncache is enabled, Cinder can seed a local cache entry on the\ndestination pool after a successful cross-pool clone. Later creates\nfor the same image on that pool then use a same-pool clone instead\nof repeating the cross-pool copy.\n\nPromotion is opt-in via image_volume_cache_promote_across_pools\n(default False). Local cache rows are not seeded from a remote\ncache volume unless that option is enabled.\n\nCache-entry preparation uses the lock\nimage-promote-{image_id}-{volume.host} instead of the previous\nservice-wide {image_id} lock. That lock-name change applies even\nwhen the option is left at False.\n\nWhen the option is False, the first lookup still uses\nclone_across_pools so a remote hit returns ASAP and the clone runs\noutside the lock. When it is True, that lookup is host-local so a\nremote row can be cloned, the volume DB row is saved, then a local\ncache entry is seeded. A failed promote clone does not look up the\ncache a second time. Glance cinder-store image clone is unchanged.\n\nImplements: blueprint per-pool-image-cache-promotion\nChange-Id: I4941197a2df741f80c317fb9e270f29f6fb085b2\nSigned-off-by: Jayaanan Borra \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/844c4469326682859f62678ec597e84aa51ed272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/844c4469326682859f62678ec597e84aa51ed272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
