)]}'
{"id":"openstack%2Fcinder~988742","triplet_id":"openstack%2Fcinder~master~I151ebb3e74826cc65829a12599ec00e0bb808168","project":"openstack/cinder","branch":"master","topic":"bp/dell-powerflex-nvmeof","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-07-01 03:27:19.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2026-06-30 12:57:25.000000000","reason":"removed on reply"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-07-01 17:18:57.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_36725\u003e by clicking the attention icon","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-07-01 18:31:27.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-07-01 16:34:29.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-06-30 10:14:50.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}}},"hashtags":[],"change_id":"I151ebb3e74826cc65829a12599ec00e0bb808168","subject":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses","status":"MERGED","created":"2026-05-15 10:17:06.000000000","updated":"2026-07-16 13:24:41.000000000","submitted":"2026-07-01 18:31:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":35,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"988742-bp/dell-powerflex-nvmeof","meta_rev_id":"ec3e24dd40f8ecf515d0004dfe25c3cc0d5ff121","_number":988742,"virtual_id_number":988742,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-01 18:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-07-01 13:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":2,"date":"2026-07-01 16:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-07-01 16:34:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-07-01 03:27:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 10:35:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-05-15 10:35:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-05-15 10:51:18.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-15 10:52:08.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-15 11:03:22.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-15 11:16:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-15 11:20:09.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-15 12:46:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 12:46:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-17 04:43:38.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-19 20:05:38.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"},{"updated":"2026-06-30 04:43:46.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-30 06:36:26.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-30 07:26:03.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-30 07:53:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-30 07:54:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-30 08:03:08.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-06-30 08:32:37.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-06-30 17:19:08.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"01a694c4176ddb24e7bea17ac0908aecc2209383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-15 10:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90f55b54685fc778b028c7be2e3b04681f8a508e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-05-15 10:35:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9cd65b7c0d892fc98af7608e2b4a0c6c98f5612","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-15 10:51:18.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/988742 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"053243b87abc9e26f38912cf3260ec434fd1915c","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-15 10:52:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/52500dca96894b1bb994f255ee3e8140 : FAILURE in 34m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988742/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"423abf08ac8125f02a4bab053661c0e36e10cd91","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-15 11:03:22.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988742 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3d2587ddf2c7c5ce2a99b451bd1b4ff629578010","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-15 11:16:00.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I151ebb3e74826cc65829a12599ec00e0bb808168 .","accounts_in_message":[],"_revision_number":1},{"id":"02cc5ec6d904e9ce17608f847ed120c5a2301468","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 11:20:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3975/.","accounts_in_message":[],"_revision_number":1},{"id":"dcb5bbb53651532417f2fc8c40bf9750a1e472e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 11:36:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4151/.","accounts_in_message":[],"_revision_number":1},{"id":"28e09105217a81753afaafe8b68975490fee18a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc119c24c724f4cb62c8d91d33fdc83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117bf428976049318e594ad3fe7ba049 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6dc80ddf4af419b86e7a92484deac09 : SUCCESS in 10m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7746c4f4b39e4b8fa59dc43382e71e23 : SUCCESS in 6m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eec4c8cc2e3b474eb50767ad4896993e : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dfecd1dda34d699132eab90167cf80 : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2ee1298f9bbb49b5b13c0265ca76123a : SUCCESS in 1h 10m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/701595f632af480ea60ddcdbf3174c15 : SUCCESS in 43m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8719bbda06394c04bed63a52302d20b9 : SUCCESS in 1h 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa21b870a8044becb87e615204ff051b : SUCCESS in 1h 07m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cee6929ece949918d8101c1e1f2c54a : SUCCESS in 4m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/64f20bbb1b654794a721302418f82141 : SUCCESS in 24m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8877e0a46dd44972ac474d9649711d21 : SUCCESS in 2m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e5b49f1680674277a5e42e388662c41b : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a67b76b7a25e437cad705113cf70dba5 : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/169b9e95da664685bef70fb69a5e0298 : SUCCESS in 4m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b71c7945cdfd4b2fba6889eeefc87776 : SUCCESS in 8m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4c8a3b732d1436d88e193624039c186 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0eabaeaa45e44e349dc2c84dace263e9 : FAILURE in 3m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c9ca96148c4410cbf194423cfcd1f42 : SUCCESS in 1h 03m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dfdc74957c7740ec8e7aee6b0f073b7b : FAILURE in 2h 27m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7549df518c5541e2885ca7933baf093d : SUCCESS in 1h 45m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1388d480762c41fbba562847afcfb0ba : SUCCESS in 1h 58m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e3dda3f9bbdb434492c212b260675b24 : SUCCESS in 16m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a9269521d41f4fcb98032bf1d8f33d82 : SUCCESS in 50m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2437fd7324854bfca86f3c3575d57070 : SUCCESS in 1h 35m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0502079df8994df8985bda192cfb86d2 : SUCCESS in 48m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e54ed8d4b22344a9a9d6e0a7e3351cbe : SUCCESS in 2h 04m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2fc063139104bc6818135acbc9c9464 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c2b9033edf9fe9bf5cc8e3102696e00f2aab7089","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-15 12:55:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_05_2026_12_49_17\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"29f6fc0fb96095651e9216301a656a1115106b9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6f611d339b34bf8822855c32280bf59\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/316c640964a54daeb287dce0fd309212 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b166353d2644b7c85b74a7e6c50fef7 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c58a79140ecd416f9c0671813f428a29 : TIMED_OUT in 1h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb4123a561edde73143bbde5200836f0740552b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-17 04:43:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/9e3e6f1f84684cae9e8f0b65660a7641 : FAILURE in 28m 48s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988742/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988742/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"0a14fd71f2ff49789996426857f9bf3906aff1e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:38.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"b5de973439d36ee04a10716addf9e8d02b805b7a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-26 00:20:05.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"459f27bcfb63a135f6cf16aa1c56784f67d85ebc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-26 00:22:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6fdfaa88dbf241c918168ffa0a6419d712f618","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-26 13:23:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1c29765ebeec2c7c78d477a911f1531c673c0f5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-26 13:34:31.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a742d01b9782e427be8dc61fbde82661e24e8676","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-30 04:43:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6256f6b4267c276ec95c9ed17b5d8d03e2a47aa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 06:13:15.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a6ca365389e1a45b84a5a506e6ae79ed9e988d76","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 06:17:01.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d70f8c4d12822304002c5be5dc3a4b1ac2687c34","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 06:23:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1d0acb3cd49ab4d55e765d3f36a5919f43a2e04","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-30 06:36:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/94e75e7df23944748a2e92b63d3fde48 : FAILURE in 22m 47s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/988742","accounts_in_message":[],"_revision_number":2},{"id":"751ebfe2b7ba51dceb96fc028131a9e2dd8b5dd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 06:36:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5623dc463a84315b9a9dc4e4b9712d2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e51a9ad067dc4033a3db1b9ae09ba75c : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ece8304b1caf420da76b197c4b796f57 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7df1b1a23010472795b384d109561ab1 : SUCCESS in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52e74339cc8f8bf4884afd690d0ccbb0afdf480b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 06:45:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I151ebb3e74826cc65829a12599ec00e0bb808168 .","accounts_in_message":[],"_revision_number":2},{"id":"ca3b8e32ef018f9c64a5486526a89a2e20c55f71","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-30 07:12:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/7b04a5680f7d4f0b86ab0e8a7f4f7e01 : SUCCESS in 34m 46s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988742","accounts_in_message":[],"_revision_number":2},{"id":"782695aed13ffdee044961a877f2c6b7cd56cd0f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 07:19:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d30332bacbc4cfb0260bf3525388af17030205","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 07:26:03.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/377 : SUCCESS in 0h 42m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988742","accounts_in_message":[],"_revision_number":2},{"id":"cc8e07f160348f97d8773ad2ad4425c20287ff7f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 07:26:21.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/283 : SUCCESS in 0h 42m 3s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988742","accounts_in_message":[],"_revision_number":2},{"id":"85e5af7f060e1d81c3b8006a0f8918a433ded8cd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-30 07:53:02.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a0cfeb8423914727abbc079f2b73e502/","accounts_in_message":[],"_revision_number":2},{"id":"76d4f9242581c7f6f0b0311c5f14b1b61e33e7d2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-30 07:53:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a0cfeb8423914727abbc079f2b73e502\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c0cd1340aa542ea9f4988b105c8908b : SUCCESS in 39m 38s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3f7571975142493fba95beec62b93af6 : SUCCESS in 36m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23e17d71878240f5b6a3ffd8e6fb65ae : SUCCESS in 40m 43s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b995318a988647d7af8fb31cc57a0a2c : SUCCESS in 33m 44s\n- pure-ci-buildset-reporter https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5bab3c34da304ad198285babf60ed805 : SUCCESS in 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1391daddab40b941280806b3491204b28f769a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-30 07:54:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5788c2b19ce14cdabaf99f08efaf2ea3 : SUCCESS in 53m 45s\n- cinder-storpool-tempest-iscsi-multipath build/d2f155e2aa054c97bc070900039e2580 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":2},{"id":"44b43e15a26555ab49624352cee67fa3419e3a37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-30 07:59:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4573/.","accounts_in_message":[],"_revision_number":2},{"id":"7a0287e1a4364480acdfa875eba7a6c6ef09c06a","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-30 08:03:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6d40e8fa3cb3cff288318075b91cc16fab7c7bf3","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-30 08:07:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9dfb57e45cb8090f3383fe24be96f4a658146291","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 08:10:27.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/648 : SUCCESS in 0h 40m 30s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988742","accounts_in_message":[],"_revision_number":2},{"id":"37826e11b95f578095e60f853c68d5417b2d0954","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-06-30 08:32:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/b858dc3d572c4503b399efdca7cceebe\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/3b3137f15b6244b4b2e283efbf9379b6 : SUCCESS in 1h 31m 05s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/bf35c0cedb534acd87b13b90b97dbd9d : SUCCESS in 2h 08m 13s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c848e42c7f73c59fb07501fe3431202b9243297d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 09:21:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc2e11100a5414f8e645c2b76b55226\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffe13e312ab4f65bc939f6e86ad0475 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42e80f0ca815444584085424b2449ae9 : SUCCESS in 8m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c597b58e9af243e4a4ecd46e1340e3b7 : SUCCESS in 11m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/967fb62a620447ef863ab5f1b8e452d3 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd694249e56e42749de595048ae3fbbe : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/e378b9d2a2a94529b3bb442db3f26bb5 : SUCCESS in 1h 04m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8791a8bf7bb04e9fa029d007fd152477 : SUCCESS in 1h 04m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d55e849f79347439e1420c5254f6347 : SUCCESS in 1h 29m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1492957e4314e708c2e0304f956a610 : SUCCESS in 1h 05m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52de7dcd35ee4db7a96d585ac73fa9d2 : SUCCESS in 9m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/25e76e2cd3514c26a3e38f5f7bd8a78d : TIMED_OUT in 41m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f486c77035c04c1a9249c1dfe44fdddd : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0f08357e4024265be26a4fd3370db65 : SUCCESS in 6m 56s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/85c54983f00f49d5bee2f49cb99e8972 : FAILURE in 9m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7178df67a44043d8ab94565929b21a62 : SUCCESS in 5m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a4f68a836c1a4ce2b516db42c126d69f : SUCCESS in 4m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0200baefbb904e7280de05ab6eb33685 : SUCCESS in 6m 42s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/619013d476084e52b6e6564cf450414a : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9895ad54170432392fdb63265137112 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c122fa46a6784dec9d7599fa45456dfc : SUCCESS in 2h 05m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4786c010a8e64bffa626178a61d0dacf : FAILURE in 2h 25m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/04a131ae87944e858eb56888334e1394 : SUCCESS in 1h 41m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/980d64164986423d9d7c031ec9a83d19 : TIMED_OUT in 3h 06m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d3dcb224037546d79715dad59a39b176 : SUCCESS in 1h 54m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/662fddfd3f0f4f2e98007e92ac335282 : SUCCESS in 19m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9ec0ec3a313e4504a085d50b41792cc7 : SUCCESS in 59m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e1a04b08499a4af18bf85692f1559142 : SUCCESS in 1h 53m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c62869866a01496cb95a8ff4bdf0c134 : SUCCESS in 27m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/36e0260881c44754959785bd9efd7e0b : SUCCESS in 1h 43m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58fa2e2d98764d46b98edbe8b6852d93 : SUCCESS in 55m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"7d3b81ba3cefc5a3f91faa8854743f002f17109b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-06-30 09:29:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/3b107d9222004f199b70e7fa12425633\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/c88526fe1cc344d389421270580bf015 : SUCCESS in 1h 37m 34s\n- powerflex-v5-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/c3234ce9723c4b118cf676059c726a52 : SUCCESS in 2h 10m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"94f5742fd5e234c3211e6a8b592b85a2648a8e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 09:31:34.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e1a7261d14ed5905a926141e288838816b60f760","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-30 09:32:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_06_2026_09_24_36\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"bb63fd9634f0aa6602993a8c197f123dc8ea31b3","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-30 09:32:25.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0debb4e02aa6b88993407b97859eff38ca5afe70","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-30 10:03:59.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I151ebb3e74826cc65829a12599ec00e0bb808168 .","accounts_in_message":[],"_revision_number":3},{"id":"0874389e90820e587b0c230cff3c27e34510924d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 10:13:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b53106152f4451a7f2382ca147abf8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1ef7ccd7a6a49a28493a27a3cad8082 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d366ac4e191e4b1ca937a31cf51417a1 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/32913be0ffa74b29a44f403f02229946 : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"591749f468c3165d1944f75626407a6c80d11048","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-30 10:14:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"91121c5e450739aeea98e4bbe2948f55f9aa5914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 12:57:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a311f3fad1643cc968008b888742b8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af0a8c28938548d997e2ec133516e2ed : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f0b96ebd724f78b212a1192dde1f78 : SUCCESS in 6m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c9d566e6ebe4cccb23ee522e065906d : SUCCESS in 11m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14c23ed46d404c0e8e3fbf54fda7a449 : TIMED_OUT in 40m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350a3f4082f7425c82d6056b496eb8d3 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/13bab9f20c394f588cf72fb3c8b943d6 : SUCCESS in 38m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b5d0615c8d04c83a4ff56e1f74ed502 : SUCCESS in 59m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c90640b53f44384a42df376fa7d0882 : SUCCESS in 1h 38m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01b800e1f34241f698a0683034616e6f : SUCCESS in 33m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8310b429f1984b2595a23468e9ceef5b : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b30fc666bf4045b58b17a15752ad8a90 : FAILURE in 25m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1868afc09247452f89aaf1193dbda74f : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5d44db8a19674da59d7ecdb3ce6875b2 : SUCCESS in 6m 15s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/81d3b24dc1ad442e8d4253851d583bf6 : FAILURE in 9m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b33cdae80d394867b770fceda4a18cfe : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c315171497994e3592f3b860d8008b0d : SUCCESS in 7m 44s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d32694aae84c425abd16c45d59ce1278 : SUCCESS in 9m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/18a50f4c4d734d62be9adc2e6f9f59be : SUCCESS in 38m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da7d91484d104972bd1c2a0122e8b2f8 : SUCCESS in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6c6293217caf4acfaba52230dad94d6f : SUCCESS in 1h 29m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2bf446789d4741e3bc1af8f394346839 : FAILURE in 1h 54m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3567c0ff9c264907a6873d3791674621 : SUCCESS in 2h 00m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3b8e50c1db5a4192b16a304e6eb9124c : FAILURE in 3h 20m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2d4e25a942514179bc6cb37b5f9c06ff : SUCCESS in 2h 02m 41s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2fdd4ccb43e649ba888e2919c73de657 : SUCCESS in 21m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/aef1a190973f433cb2292aa5accbbe3b : SUCCESS in 1h 16m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aedf7e102e6e44c99dcddfc1a9241cc2 : SUCCESS in 1h 40m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ccd43dd78fb74de59e96f854c732216b : SUCCESS in 34m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/77bfcb7126e3494485fe1e72b7a2b8a4 : SUCCESS in 47m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac043e7bcb54f8b80912e377af83eef : SUCCESS in 49m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"0a966c60c8cdf1af31b95d489caaef5fef130804","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-30 13:08:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_06_2026_13_00_41\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"5ffd294de9365435595e7089a9b0457180692a09","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-30 17:19:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-988742-2-fc : SUCCESS in 1h 6m 5s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-988742-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"cd2f0c01b37b5e7e11386c6079db827eeb880e6f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 19:22:39.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/301 : FAILURE in 0h 30m 9s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/988742","accounts_in_message":[],"_revision_number":3},{"id":"41f18c6651337930b046fd22491db5e769674c55","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-30 19:51:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/81cb7b2ec3d64191aac470801dcf07cb : SUCCESS in 59m 18s\n- cinder-storpool-tempest-iscsi-multipath build/ffeb79105e9548459c88666182dabb2b : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":3},{"id":"3d7bc18b7abe597a0c0246eafdd3b678039a4219","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 19:58:07.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/395 : SUCCESS in 0h 40m 43s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/988742","accounts_in_message":[],"_revision_number":3},{"id":"978ff3cff0eba52af054e03d60777ee3da7b494a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-30 21:11:00.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/667 : SUCCESS in 0h 42m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/988742","accounts_in_message":[],"_revision_number":3},{"id":"b574cd87e534c2e8509118495aba2178f9b9b727","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-01 00:31:14.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-988742-3-fc : SUCCESS in 1h 4m 39s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-42-988742-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"214c293c5a9cbb4c5f00d1cbe546cec0abf32b2f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-01 01:50:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4591/.","accounts_in_message":[],"_revision_number":3},{"id":"e5c914e618079b27022b5690495301405fca6551","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-01 03:27:19.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"791b08e515b8d75a45cb1eaadb5a2542c2817733","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 04:55:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8c1f4e3884a08db5458ff3f8ab2c69903e3c23f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 04:56:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90f6033b3c69f2b423c93741fa62a496a2207b9f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 04:58:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b7ed978f1de3601ccb76c4c616feef7f4d60cf0","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-07-01 07:05:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/f4a3a5184ff44ec3ba8c106cb3e48403\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fcbfcf03cfb34891953614fbb89d9204 : FAILURE in 2h 08m 33s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/754e93eeedad461795dab8a5a05e48be : SUCCESS in 2h 01m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"bf85ce9cb64651e90163890ff299d29e86e80278","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 07:15:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27b40874b08f5eff0a44cc483b70b4a2511b2457","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-07-01 09:09:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/0153c95c64874c7baa5de9e642d4574d\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/aa5d7303c7504c73a6edb1af82442b51 : SUCCESS in 1h 37m 07s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/7716e43ac988494fb0d323d79baeb91d : SUCCESS in 1h 53m 26s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"51d88bdefa00159458205e1f4a6f93f44f2140fa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-01 13:20:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d4fb64c3d8d98a6da39bb1291f56dce14bd697d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-01 13:22:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b153e8eedd5e47b7f2506d96c6448d79fe53130b","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 14:30:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1ed38ed20bbf6bf2e78435a0e6d4823b68ca88e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-07-01 14:30:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35ebd6bea191a87d2638ea778fc17ab7ee09f09f","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-07-01 16:29:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/458dad1323c44d68a50f5a18af91aef3\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/5165533d01da4cb7807717c1ea6058e8 : SUCCESS in 1h 47m 36s\n- powerflex-v5-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/ba3c70c03d7e45818e2b8df40f39f18b : SUCCESS in 1h 58m 42s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5121e99819995d62de7603f76426136b93c8d94f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-01 16:34:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1adf131797b80fc1e8c40f175fb2cb91bc73bb05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:35:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a264e822bc728acfd643077f0016f23117f3695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:31:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ec2fccd38842489fef5b1670de02ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eced93160afb4788b9cd7f81f49e7cf6 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d19e541952774c5a85887dcad8954a4f : SUCCESS in 12m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4045d7cd6734ac2af7284709e005bcd : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925f19af5c114ab6b24fed9b8321c6e5 : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/066944ed4d1e4ff9adad206b8fcc2a85 : SUCCESS in 32m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ddd6e8c016543039fe92bf14aeebbb5 : SUCCESS in 1h 11m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/821a2683c95f43f3b904fa0dcdc1a672 : SUCCESS in 1h 51m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b221f494bee94f8588b67ae079505c5e : SUCCESS in 1h 02m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eebc25ada3b741b1affbea5cdaeb5c99 : SUCCESS in 9m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e54ceb9a4cf34de78756bffc1852b0c1 : SUCCESS in 1h 16m 40s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a06910a45f34fb3abfda1270da8ebd6 : SUCCESS in 1h 51m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4909ed23bd6c4c75b1adb53381ca7509 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":3},{"id":"077c4a046ecccd7831b0b2836ca2193296d34971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:31:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c0c4e7ceb25279ecaffdc2cdf6e7d59d2f26105b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 18:32:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/953b0eb0f58444eda607b8e9427dfa88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038066e0b6dc42e09c6aa16751ff527d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"57cf2d6a138e9592bc83c07f590559463c8ba28a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 09:49:54.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4397/.","accounts_in_message":[],"_revision_number":2},{"id":"ec3e24dd40f8ecf515d0004dfe25c3cc0d5ff121","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 13:24:41.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4415/.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2a8f9546fa60bb42e3a62ada2914553ab0e42c0","revisions":{"b7c09ac13949dddbc7bfc4a9aaee71119d2d7de5":{"kind":"REWORK","_number":1,"created":"2026-05-15 10:17:06.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/42/988742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/988742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/988742/1"}}},"commit":{"parents":[{"commit":"d7515145399b45d23ff00e64b5ef9fc6ae55788e","subject":"Merge \"NetApp ONTAP: honor ONTAP 9.15.1+ default for LUN space-allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7515145399b45d23ff00e64b5ef9fc6ae55788e"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 10:12:44.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 10:14:33.000000000","tz":0},"subject":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses","message":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses\n\nRemove redundant \"errorCode\" in response checks from query_system_nqn,\nquery_sdts, query_hosts, and create_nvme_host methods. This ensures\nerrors are properly raised even when the response does not contain an\nerrorCode field, making error handling more robust.\n\nFollow up to https://review.opendev.org/c/openstack/cinder/+/912058\n\nChange-Id: I151ebb3e74826cc65829a12599ec00e0bb808168\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7c09ac13949dddbc7bfc4a9aaee71119d2d7de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7c09ac13949dddbc7bfc4a9aaee71119d2d7de5"}]},"branch":"refs/heads/master"},"d25f41ff8ad74d00f2c8b1d4370cd5363a2715b5":{"kind":"REWORK","_number":2,"created":"2026-06-30 06:13:15.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/42/988742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/988742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/988742/2"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 10:12:44.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-06-30 06:11:39.000000000","tz":0},"subject":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses","message":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses\n\nRemove redundant \"errorCode\" in response checks from query_system_nqn,\nquery_sdts, query_hosts, and create_nvme_host methods. This ensures\nerrors are properly raised even when the response does not contain an\nerrorCode field, making error handling more robust.\n\nFollow up to https://review.opendev.org/c/openstack/cinder/+/912058\n\nChange-Id: I151ebb3e74826cc65829a12599ec00e0bb808168\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d25f41ff8ad74d00f2c8b1d4370cd5363a2715b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d25f41ff8ad74d00f2c8b1d4370cd5363a2715b5"}]},"branch":"refs/heads/master"},"a2a8f9546fa60bb42e3a62ada2914553ab0e42c0":{"kind":"REWORK","_number":3,"created":"2026-06-30 09:31:34.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/42/988742/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/988742/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/988742/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/988742/3"}}},"commit":{"parents":[{"commit":"85611cbc18efd44f27cce2b5dfb8b2d156230ba8","subject":"Merge \"Storwize: stop vdiskcopyops loop racing sqlite in tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/85611cbc18efd44f27cce2b5dfb8b2d156230ba8"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 10:12:44.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-06-30 08:46:45.000000000","tz":0},"subject":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses","message":"Dell PowerFlex: Handle missing errorCode in NVMe-oF REST error responses\n\nRemove redundant \"errorCode\" in response checks from query_system_nqn,\nquery_sdts, query_hosts, and create_nvme_host methods. This ensures\nerrors are properly raised even when the response does not contain an\nerrorCode field, making error handling more robust.\n\nFollow up to https://review.opendev.org/c/openstack/cinder/+/912058\n\nChange-Id: I151ebb3e74826cc65829a12599ec00e0bb808168\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2a8f9546fa60bb42e3a62ada2914553ab0e42c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2a8f9546fa60bb42e3a62ada2914553ab0e42c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
