)]}'
{"id":"openstack%2Fcinder~988746","triplet_id":"openstack%2Fcinder~master~Ibf0eb285aae37a456a367e7ec749543e7a743733","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-06-11 13:01:35.000000000","reason":"Change was abandoned"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-06-11 13:01:35.000000000","reason":"Change was abandoned"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-06-11 13:01:35.000000000","reason":"Change was abandoned"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-06-11 13:01:35.000000000","reason":"Change was abandoned"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-11 13:01:35.000000000","reason":"Change was abandoned"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-06-11 13:01:14.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}}},"hashtags":[],"change_id":"Ibf0eb285aae37a456a367e7ec749543e7a743733","subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","status":"ABANDONED","created":"2026-05-15 10:34:27.000000000","updated":"2026-06-19 20:01:25.000000000","total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"604214fb8cc9c10597061149f81f268510db7e1c","_number":988746,"virtual_id_number":988746,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-04 14:33:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":-1,"date":"2026-06-09 13:28:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 11:23:56.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-15 11:33:12.000000000","updated_by":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"reviewer":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"state":"CC"},{"updated":"2026-05-15 11:34:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-15 11:48:48.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-15 12:25:50.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-15 12:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-15 12:50:22.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-05-15 12:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 12:56:06.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-17 05:18:13.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-27 15:32:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-28 06:23:41.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-05-29 17:17:33.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-02 10:33:55.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-06-02 10:33:55.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-06-02 10:33:55.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-06-03 15:01:41.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2026-06-19 19:59:57.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"ea8adc4ffd681d81c64b4d7b3ce2a4c5409d4787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-15 10:34:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"009a516559cbd5708532db6ead67d11fec9a97c6","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-15 11:23:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/14582d286c76471186ccb5863691dfab : FAILURE in 30m 26s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988746/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"f72d898ac215b16c41bff145f087b063fff1d039","author":{"_account_id":38914,"name":"Palak Agarwal","display_name":"HPE 3PAR Storage iSCSI CI","email":"palak.agarwal@hpe.com","username":"palak12"},"date":"2026-05-15 11:33:12.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-iscsi-driver: Logs: https://github.com/hpe-storage/hpe-3pariscsi-ci/tree/main/new_logs/988746 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"83d25a36f2bfd10d0c4a1772017723478fb65349","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-15 11:34:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf0eb285aae37a456a367e7ec749543e7a743733 .","accounts_in_message":[],"_revision_number":1},{"id":"b30264b31d2b53ffa07cf76891f9bac3323e570e","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-15 11:48:48.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988746 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cc2760a60a5dca9d486f6b70b30230174fce0bb6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 12:25:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3976/.","accounts_in_message":[],"_revision_number":1},{"id":"3e24ef235b4fa72835aef5e167ac70021f311dd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:46:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12977b3970ba4a81b8ea5fb3417d372c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52c700c29e7d4de5afb961d0ec63f8f4 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05db8f86fca44917b51c4444ff81c6ab : SUCCESS in 19m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2a5c7856e51b4937981477fd010cd06b : SUCCESS in 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0084fc21982b3535792bb53d4417101ad98e8886","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-15 12:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4152/.","accounts_in_message":[],"_revision_number":1},{"id":"d93308fae6d46b759e64108359dd1aff99b806cf","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-15 12:50:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"422990a88d5c78f1d4af19a8f588e65a33c4ce1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 12:55:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d0471c02be4810a2b4754c945699f0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f4cefede8b6240f5bdf0c8d9d1d5770b : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8105885251487db4cb66feebf3bda3 : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65ab29d78fed4d458bb79caae3cf1d34 : SUCCESS in 7m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd2d352db0d640dabffb486bfd51befc : SUCCESS in 9m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4605da6d95424e418d94940e47cc754f : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bcae28ad38a4438a73b59ba1a8d671d : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/54e7067212674b03959c9f77c78318ed : SUCCESS in 1h 08m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d1274216054742d3baa44e4427d2a46b : SUCCESS in 56m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bf237050e404afab0fa04f98610d014 : SUCCESS in 1h 30m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6413d1064fa147d5a0275987dfbf1044 : SUCCESS in 33m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/319edf6fd7454ec1ab6c5507c8d53b0a : SUCCESS in 10m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a308138716a54843af0357e0c3a542da : SUCCESS in 25m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8be47f4b90e24bf5b09055078d90ab8b : SUCCESS in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2660a4b5d854dcabd16a72820c7a569 : SUCCESS in 6m 32s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/0e57014fb60541a28127d61100ec990b : FAILURE in 11m 48s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/dd74624532554d659b2f6935e06584a4 : FAILURE in 13m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a9744983235746a0a8baed094accd0bf : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8bc24cdf16d84af6bd03d785ef2a11f4 : SUCCESS in 5m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7d1cf68e713440ed943f67c45e1d6a49 : SUCCESS in 8m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cf329e53de374a228ed497aa535cb457 : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2e3254189564e1abce115da179abaff : FAILURE in 4m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c579748df8e54b0c8f675ae4441c7672 : SUCCESS in 1h 56m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf982bffc95b45599f98429a36d4c872 : FAILURE in 1h 39m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b998343443cf426eb47c181123f8410c : SUCCESS in 1h 49m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9efbd518844b4258b6047463134229c8 : SUCCESS in 1h 54m 35s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3674358b4eab4b65aaa2439395474612 : SUCCESS in 43m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f28ad9a27ec045b1beb628c72719dd5d : SUCCESS in 1h 14m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b9810c240f35468f94b084914dcb94cb : SUCCESS in 1h 39m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5f3935dc82c04d8aba67afff083e08ee : SUCCESS in 44m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70bef041661845e392b9bc74182a4a8c : SUCCESS in 1h 07m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55537293dccb4c179c0ab6370361204c : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"549e4aed8957d50b0bf87c00adc9b43025874fbf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-15 13:05:40.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_05_2026_12_59_08\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"8f1d7c53bacbc035e9b2c00f97f6c8879af3f8b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-15 14:23:54.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"e23d92af1147e7aa3cf8c376f3997cfcbf675c46","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-17 05:18:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/69da7968028a40d0957a5083f563dc7b : FAILURE in 29m 31s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746/etc/tempest_log.log\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746/etc/tempest_log.log","accounts_in_message":[],"_revision_number":1},{"id":"ebef08616c78966bd804a653fbbfa532d0575e24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-27 13:06:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b617c29dadd638074c31c8596168d8f2f906960d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-27 14:21:31.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibf0eb285aae37a456a367e7ec749543e7a743733 .","accounts_in_message":[],"_revision_number":2},{"id":"888dbebd22cb71c627fa900535608ef4710cbfd4","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-27 14:25:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/75aa15b2f56842599edb3f98b1e90bee : SUCCESS in 37m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988746","accounts_in_message":[],"_revision_number":2},{"id":"e93fb843585f8a0b6e32fec33555d5f5ee08cb2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 14:34:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d89235e6a694e63b691260df084fff8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57dc095060b14b7f95c7d667c3d4f1d8 : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0de2e2d753434a8b8464bd2fa0fbcb82 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/75c07e876bb74307bc04c577d381f3b2 : SUCCESS in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3abbd19b4177085de6331ba355b54d2da5cc48b7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-27 15:11:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1aef7815337440b1af6741354fd833a4 : SUCCESS in 39m 59s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746","accounts_in_message":[],"_revision_number":2},{"id":"1ca64394b9d2002c2c7871fc93117f8e9386605b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-27 15:32:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/447ec7b978464889a31bb03917c525a5 : FAILURE in 57m 17s\n- cinder-storpool-tempest-iscsi-multipath build/03ecd1394fff4016baee2381fd6a5d3b : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e44e01c888ed12a123d75123d29085976740da07","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-27 16:39:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4253/.","accounts_in_message":[],"_revision_number":2},{"id":"e227372b09d5b08191e2e6108f0710a7afa688ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 16:52:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f0c9ae7261a4d0299cd91328a4ef0d2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebae597854564044a1eabf5f97879359 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0936134ac84239b777b339e51737ca : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da31ce19d3e446628ab1c997597fee11 : SUCCESS in 9m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c984054e993d4a68ba602716853ab6b2 : SUCCESS in 12m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e46cc3b949f04f688dfe2345c8c4f0ae : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62992fdfdf2e4107bc008d0696ec5ed0 : SUCCESS in 17m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/472b0df06f974164be5a3cc349ecb078 : SUCCESS in 1h 02m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6ada5d4a1bd4165ab304cb0b782d20b : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a31426ddeab407fa230ca0dbdb2f51e : SUCCESS in 1h 43m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37c272140fbc4d8f8763938cef885743 : SUCCESS in 55m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19794e703486481cb48d62db46b4df8d : SUCCESS in 6m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2df7f3ade8e04bfbbf0251a1154a38c7 : SUCCESS in 18m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf07db29c1e64e66b22a18821e31332d : FAILURE in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a01b6236a4ec454e977e120093624172 : SUCCESS in 6m 42s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/88f78a99255347bea13a00e782e14be1 : FAILURE in 8m 19s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/7222d6a9d050441ebf0b2a6422ca7af1 : FAILURE in 18m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d0a3662b283422ab45754cd01063988 : SUCCESS in 4m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a47e9c709103458a820091f387eea824 : SUCCESS in 4m 47s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f02be10a5fe54498b54f55caac4b154a : SUCCESS in 9m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e8b882313dae409188e78cb6d7ce0f23 : SUCCESS in 43m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1264b404ab7346949edbccc99cce90b8 : SUCCESS in 7m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f223094ba0c0435bab36ac7e6b375787 : SUCCESS in 1h 14m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4711a446322e4671873114643291c3d7 : TIMED_OUT in 3h 41m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/897df6822dcc4a51986dc5b05e6a8cdc : SUCCESS in 1h 49m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f0d6701db78a44e7adaf3aee8a2927d9 : SUCCESS in 2h 12m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/23b28febe7c84f4e8bc3e2ec2941a896 : SUCCESS in 23m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a879396f7f24e56ba855b84fdc314af : SUCCESS in 1h 16m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c55a7b9b598480ea258fb9076650488 : SUCCESS in 1h 40m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/51cac3077e8341fbbac8f86e97a1d6ea : SUCCESS in 35m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2672aa72e74494cabdb6032f3a4a192 : SUCCESS in 1h 09m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22e345a286d94f5eb95db764da783fa0 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":2},{"id":"31c15041dcbf592fea50e73fd54c01df1888c06a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-27 17:47:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4077/.","accounts_in_message":[],"_revision_number":2},{"id":"4894cc8d97589d13418f92bb5ce4cd31e328b714","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-27 22:13:54.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988746 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e88bc33e00b36f3bce1139b14c1f545b4f603ae8","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-28 05:59:00.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c07104d4151d21a54fede6c32f76c583b0116057","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-28 06:23:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988746-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988746-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ff57f3aedb64af81058cac83e2f2f6fdf667957f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-28 06:59:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ba3971e7799a4363bffbe45a847f796b : FAILURE in 58m 03s\n- cinder-storpool-tempest-iscsi-multipath build/5e58dd7768834f7fb6657b6ecf88f682 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":2},{"id":"750dee3db9f538f5f9f520476dfb1173f9ad86e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 07:42:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aad6bd1728048aab7b747d18d3a983d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce8c3425ce1d45858f1495e2add39040 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9e51d377117468d8956f610abcf7b24 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bcccdf75384c408f937765bf67807e62 : SUCCESS in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc2673156ed514b37f90efeb3b6b34ba00123710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 09:28:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18a7e6adbce34b27a73d6040715e911b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49ea4954f17f4c10a7073749ada9597b : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4cb670bccd4503a88b43459eac88ea : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/728caa8d601446fdbc409b26df516240 : SUCCESS in 7m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48a80492c2474f099dd2f0f9a82e6f0d : SUCCESS in 10m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f127dbb05ae2445182d15161334f2109 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ead42a48f094a51a8c922898e8e2b1b : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/daa58c22b52e496e897e48e4ea2d2ab1 : SUCCESS in 56m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f08f4995f9d2460192393ee35d1bca87 : SUCCESS in 1h 13m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb661259922441ce851e548982d83775 : SUCCESS in 1h 26m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff0873d33b3a451395783b94760726d5 : FAILURE in 43m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12fe8fd7d1c545a699d9db07bd85c0b8 : SUCCESS in 8m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e4e615540b4c427ebd6d2cce9b722f92 : SUCCESS in 19m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e4a4ca614cab4223ba6c05cd39db255e : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/234fe516b31c4b299ec0bb96ae20d11d : SUCCESS in 3m 07s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/87ee7a3b087c498c8c703924e994491a : FAILURE in 13m 36s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/ba37b805e73845a2b36788eb032f9a2c : FAILURE in 13m 22s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91a1e41f80104f069372280a3c6a4436 : SUCCESS in 5m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ef7cfab6d86843dfa7d36e81b1c396fc : SUCCESS in 5m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f6f32e7abc3f45f89f5b19568fd1f0f2 : SUCCESS in 8m 55s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02a18f07dc24480ea11bfa6c8cdc68d0 : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47ce755a1858421a88be47754f1d1bf1 : SUCCESS in 3m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a949390c780c42c08e929762482a6c60 : SUCCESS in 1h 09m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00f88d17f02549e985ee6f998fbd3516 : TIMED_OUT Host unreachable in 3h 23m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94864134487a4d99a5de4a78140f728b : SUCCESS in 2h 26m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/97395ac93fe14f50bdbebddf442d02c2 : SUCCESS in 2h 01m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/eed72dd67c844fd48887e6debebbdda7 : SUCCESS in 35m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7f36f1ff47e49258f64a1854a44378f : SUCCESS in 1h 14m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c0cd1d91440f499b87c72d794c5ce14a : SUCCESS in 1h 28m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1347d578dedb4a239cfa55fa42ab9a82 : SUCCESS in 26m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cafaffc5c8694369995cd685283ab06e : SUCCESS in 1h 18m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87485582f3f8407aa8146f450dfbf13b : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":2},{"id":"7eefdcc5f3e5967ccfe02e3e52e914d49248ecc6","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-28 09:31:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44d641c4e62a4b936f0b158b6e46b910d5b9283e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 11:02:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1881bb7ca6497ab82368509bb72bc5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8355d914c784101b32cdd078b7950f7 : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/12ccd85d5ffb4eed8e9b0d787a5a9f43 : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b892c2854fab4c5ba0f4b239a2716d1d : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"388464a4607d7aa15eaa16a66ee295990a939bb7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-28 11:19:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/2351099d797649a783da1d19bd88b74c : FAILURE in 54m 41s\n- cinder-storpool-tempest-iscsi-multipath build/a9f8248c681945d6b7029077bed8e699 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"82938e4ccbbf6f697094f0ebe61ee8b2d735f061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 13:19:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042bce96e71342a5bd97480389e4d024\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8fd75bd7d7c54714bbaf00bcaf46cc39 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f43c37042ef480b99ad9c66eac64516 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab9e3d137b3445f6bf12422c1fb61baf : SUCCESS in 9m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e77463c1a3db492cb3a304de15e95040 : SUCCESS in 11m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3bdacfd5144e4b7a81679312e1f3a074 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8677a982ef304002bd5b33245df1833a : SUCCESS in 16m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/fa1b37f0260f419991efe37df2130f38 : SUCCESS in 37m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cad9ed0588f421387506098853c91ed : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/243c93d1200f45fabd2b4fa757a80c89 : SUCCESS in 1h 46m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/452cc3b2ba2d4181a8fc00ed5a927c15 : SUCCESS in 33m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60f681b8eb9449209fef7db4fdfea6cd : SUCCESS in 7m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5e59b4f50b7b4f2eb980a44fbcdcf45d : TIMED_OUT in 41m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8da06f44dd1a47218723ec51292bb772 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/552ffd4ca7434db983e56b486ef2b33f : SUCCESS in 4m 56s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/b404d8671a1f40f58f916db618d87442 : FAILURE in 12m 42s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/f42f009957954022b6726074bd97cf68 : FAILURE in 15m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9b7859cf806f439c8f6c7fa85153045e : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c8d0ab25b595474abee9497e97ec6cc4 : SUCCESS in 6m 48s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/fc5c6cf368924a8b87da4416a1e5e244 : SUCCESS in 8m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/adf9e35d62b04098a21ccb11523c55d0 : SUCCESS in 41m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6ed8b671b1884422b48871bdfb6046b8 : SUCCESS in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a70fa48691af45109c009cb989ca0750 : SUCCESS in 1h 52m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d93d832b1684c54aa5d854f100f3883 : TIMED_OUT Host unreachable in 3h 42m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b49ee51864b645ec8872723aa4768fce : SUCCESS in 1h 56m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5151a695075549c0aa9cfc4c64839cc3 : SUCCESS in 1h 42m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5da45ffe9a704e6188b64c3c118e9284 : SUCCESS in 36m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ecf9f7274ed44f7483907649d28a6d50 : SUCCESS in 1h 21m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08414fc0278d4c48b94cc12fcc16894c : SUCCESS in 1h 44m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b27d71bf9de942b8be30bebc237882d3 : SUCCESS in 45m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/623dd5512ebb41e09708657590fc70e5 : SUCCESS in 1h 51m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86f1543509db45d69ee0b54668b24cc2 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4b59323dea08ca271d273fa1484ddf1a8a99da2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-28 13:37:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_05_2026_13_30_27\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"b887cc51b8dae2154d9e874f51c317b86537e1f7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-28 14:39:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ff2381ca94769fe1a17a4faf300a7ac0f4ede21","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-28 14:48:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0724057f7273091bbe359c7f46b1058f34501b45","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-28 16:38:03.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988746 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3796badca5c343853f0ac6af36d079962e8d59c6","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-29 00:32:22.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/988746 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"7388c6f7b822548547b741bd43972eadaa7f487a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-29 15:56:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c251707af22dbf8ba854bbaaeb49661ca0357aa4","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-29 15:59:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"960b6ed7bd555dac5ad3a173cb2f89b9f0d6f853","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 16:18:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/ca59ea2a55d94475953c4c310f856082 : RETRY_LIMIT in 18m 41s\n- cinder-storpool-tempest-iscsi-multipath build/a0283763804840ef88f9549e4e050115 : RETRY_LIMIT in 19m 19s","accounts_in_message":[],"_revision_number":3},{"id":"6ad291b3286b5685bf24d52f55cabc4d025c46c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 16:53:43.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf0eb285aae37a456a367e7ec749543e7a743733 .","accounts_in_message":[],"_revision_number":3},{"id":"1d46fae8cb9327ea3590de43deec2c6b90b122f3","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 17:17:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/e79bf1f1368f4a96a716f06c1ba664d6 : FAILURE in 22m 03s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/988746","accounts_in_message":[],"_revision_number":3},{"id":"86e46c3442ca591946761b468700a0c5f3d5d241","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 18:15:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver build/5f90ec91343e4407a3a8226a526ec9b7 : SUCCESS in 36m 51s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988746","accounts_in_message":[],"_revision_number":3},{"id":"0d656625c77cf7937ed0bed8b2c0941f06c49658","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 18:54:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bd3281c9012c4482b0da745ba81d95ff : SUCCESS in 41m 01s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746","accounts_in_message":[],"_revision_number":3},{"id":"0a96c504912cb81bfcb080259e806815e869ae70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 19:46:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40783d80eaa44bda0a29aecc89648bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7698babf424548f8982a52482997e6c8 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bd0c5106a343c8983c6e74fa33026d : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40cc7015ab734b3f93fbc9ade9c2e453 : SUCCESS in 9m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f939fb80880f4d85b0772a91993422ef : SUCCESS in 10m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76de1945d71648c58253fe2f7d318641 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750c918fec39460da7380ac105c3e2cd : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/7c17533f62f041db833f7aa9d6556f9e : SUCCESS in 59m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41823005d679419ca9bc58f197990918 : SUCCESS in 1h 05m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c71a0a4944044184a65a18fc8c512424 : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7e061fb6221498d819d8b44f1c8829f : SUCCESS in 49m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c3bd67c5824ab596049acf31369e08 : SUCCESS in 7m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/18f47f0ad0e44d6787c2a21d39444653 : SUCCESS in 17m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7fa5a89592e64827a80cb8bda289e3d2 : SUCCESS in 2m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b9a65e18512e4f74a95e0e1585af1369 : SUCCESS in 6m 35s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/83e2e44c166549c392ae85fb4adbdc65 : FAILURE in 11m 07s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/496d7c0822f64b6bb3872634d8eab3c8 : FAILURE in 17m 20s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68629eca118f4ff88e85a51ac79794ae : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96f3b5a798474a12a40c2ac27f974859 : SUCCESS in 5m 59s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5051fd80fccc412db6d2336385e110b2 : SUCCESS in 9m 00s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/455c8bde7259472cba73eb7833fc1ce6 : SUCCESS in 30m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a35d358c22c4c9db4ce396d3122b1a5 : SUCCESS in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93975143a44943b2a3e8f563203c257d : SUCCESS in 1h 07m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58fe0757fd1445d5b3783ca3e0bc793f : TIMED_OUT Host unreachable in 3h 42m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ba0610ec5204ed888641d09ffad7cb3 : SUCCESS in 2h 01m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c6936b72a44f4255b9893aa2c4280030 : SUCCESS in 1h 57m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4c9e9f50423845e295f57bdc0f62a60d : SUCCESS in 39m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d492cb6a2b4a4027abd528fe8e29e5a7 : SUCCESS in 42m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7a52031d15604f3ab7ef3c643e2730be : SUCCESS in 1h 44m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/80f2776464be49bd8936a80d21b38601 : SUCCESS in 44m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08a1a481f8b04200a94f91c4f049b8ef : SUCCESS in 1h 51m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65980afb7f8b437c91d2e53318f8f194 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":3},{"id":"de4270e4b474628a06eb599259f29800a9f47932","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-29 19:57:43.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_05_2026_19_50_12\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"138e179bd4becc24cc611c84818f92c3c46e8efd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 20:25:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5dd9069feee414f92fb179c66f1d80b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a65f33dd7e944dce8124fc8b5eeae73b : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aabf292435d4453ab58a842dc2bd4e3f : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/837fd0c1216a4fa3aa4cabda15d346e2 : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a577e84b1f4ecb1cffe285bddb348d771371683","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 21:43:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4110/.","accounts_in_message":[],"_revision_number":3},{"id":"09b4a35e1859f63b779a662cf23dafd4e63ff9d7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 21:49:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4286/.","accounts_in_message":[],"_revision_number":3},{"id":"08bffb9cf4c34f0941653e0b3551b4763c077cf8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-30 01:30:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988746-3-fc : SUCCESS in 57m 49s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-46-988746-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"3f9cbf8492f13d2e7b34abf2b3188b6f2e6a419b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-03 04:52:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b305bef2e0dd292a79f4a6ae8ff5afb11d0d5571","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-03 05:15:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/9a7e93bac7974ecc93904a23253ed472 : FAILURE in 22m 32s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/988746","accounts_in_message":[],"_revision_number":4},{"id":"b1fc14aca6c7fa7f60aaac17f97ad12559fd6f0f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-03 05:35:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/45460a8166e2451d8ec10e4f54fe39db : SUCCESS in 39m 56s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746","accounts_in_message":[],"_revision_number":4},{"id":"3510da3af64d4a29647c5fa2702308b7fa8609a3","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-03 05:35:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver build/b636c32a56774fbaa03acd2e1bc3e223 : SUCCESS in 41m 40s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988746","accounts_in_message":[],"_revision_number":4},{"id":"abf6e4c9e1351e39ddf3bdb832cb524e2fb02471","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-03 05:49:46.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf0eb285aae37a456a367e7ec749543e7a743733 .","accounts_in_message":[],"_revision_number":4},{"id":"5ab6d623722f7a2666039bc69774a4b6960044fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-03 05:52:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ba9100ebc46745b3b8bbabbbcddaeb61 : SUCCESS in 55m 19s\n- cinder-storpool-tempest-iscsi-multipath build/654186bfbb5146e4a156cad3604c9d0a : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":4},{"id":"df026ae47b6a5f964e2a4af690d6f938302e9d97","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-03 06:57:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4157/.","accounts_in_message":[],"_revision_number":4},{"id":"6070d1d15bc0fcb806efdfd479b3a26914c436a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-03 07:06:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4333/.","accounts_in_message":[],"_revision_number":4},{"id":"21bda76aacf3cb2792167dd8c7fbd511795b6ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 07:38:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1dc04ce439441adb0dfbb3ac4b451fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9adeefb61144c77ac34fae2bc9b839c : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b078ce64ef504db1a8154c5c1962de26 : FAILURE in 7m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba029a94b45d40a98ac9cb696adb1212 : SUCCESS in 6m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f229d3316c7c4e1585c0de883b127153 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75572681d8b4b1c88287851fb933cfb : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a34c3dd1cbf946d4a818b7b689a94b98 : SUCCESS in 32m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9197de021f844b7b5266f307692ad6b : SUCCESS in 32m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37baa139a09943678924a465f729a006 : SUCCESS in 1h 27m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0a764c1a62480381405bed61628416 : SUCCESS in 34m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7078fd80e29f4cb185bfcd77de223c7f : SUCCESS in 6m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5c938d41fda3426aa53db70a586f846d : SUCCESS in 26m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e112d8c0fdd64350adee0df6a3388ab7 : SUCCESS in 3m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22d6afd6bdf140e894a1909ab7d2d3e7 : SUCCESS in 3m 21s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/3cbf37e351b941bcb16e6fdfe6e22043 : FAILURE in 7m 43s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/83e742429d9a4c5fae00d72d2f245d6c : FAILURE in 14m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/83ef495c2c3d46869455c05743451e0a : SUCCESS in 5m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3c6198aa8ce4fdba2fc4a58d01c2218 : SUCCESS in 4m 49s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f97f8f3e56cd44e9b05b6dcdc3aec98b : SUCCESS in 8m 25s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9925873f765c434a996c428a6f26c1c0 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/65e07a0002e8490895b5f84396eddc49 : SUCCESS in 5m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0775fc102115458e88cbd085f06eb59d : FAILURE in 1h 49m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b79880dc548b40748571f1e8e6689081 : FAILURE in 2h 40m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b026bd643a554d4890c15b655c69f25e : SUCCESS in 2h 00m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6ad392e6737948209904ee41a4b720eb : SUCCESS in 1h 53m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b186f2b654b04b8da2d89fa867799db8 : SUCCESS in 18m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e424a06d79d454fa56b58468e400aeb : SUCCESS in 1h 09m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0bcd40cba9d548db80b526b6ba682f08 : SUCCESS in 1h 41m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/683482196a3b40c39158727a47962f86 : SUCCESS in 35m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/00384d6a589241d9baca5f88db90a465 : SUCCESS in 1h 42m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/754df521eaab40258cef104ba641711f : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":4},{"id":"ba597c623ca959149ed09d49e5788b4a87932336","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-03 09:11:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"260093b55d95284438e32f4a07ad89a409c9d078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 09:39:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63fa8147afd24fa4919caa9202403c55\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c188591871da450aab9641523caab968 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ffba7122dc774921bdad40b37635f70f : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9b5c1c2996954fe581b9f5d9369ae7fa : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"beba3b66d1fb37f104ec5610c3b916b719738238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 12:02:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb4a65fa552497181921a8af533c012\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80153b0a2c1b48afa276e1e8007f3a4c : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9995680c65504c5ebf4f66bb8f811404 : SUCCESS in 7m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17d23af0bfd24e7b885652895fd632db : SUCCESS in 11m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4df9eefc5dd7410eadd475d6f5755c2d : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3eb057683249ceb8017c6b2edc49d5 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/4c42789e3e0a4802a41189de561a04e6 : SUCCESS in 1h 07m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a96dad6435e940518218e6b9bb397fc6 : SUCCESS in 47m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91c777476d334b1e9d4322bef6fbc430 : SUCCESS in 1h 52m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbe53870f68b436f8e765c6335d059f3 : SUCCESS in 37m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f1e15fef8b24e919fcea0f384b98e73 : SUCCESS in 9m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8450adb8117747aaaa3e636f4cb0caf9 : SUCCESS in 21m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6766280080114d34a18cd8f6617b9f97 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e50258db9d9f4a2e9ec71ebb1ba3fbb9 : SUCCESS in 6m 29s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/3ca9953256ef41cea68aa6ef9040a03f : FAILURE in 12m 09s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/1fc99d22ff80439da74ec966bc2789d0 : FAILURE in 17m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cc3cc2b9031b4de4986b0d3f4e071efc : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6a34222c81424b939a371f675e061a6d : SUCCESS in 6m 23s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f99884be87aa4f2fa02a49d179172fe9 : SUCCESS in 6m 23s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ba72937aeab1467e9e00d8df451d3987 : SUCCESS in 45m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9714c5057de0429bb063652f1989d756 : SUCCESS in 3m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/99a7efc749ff4170a2aaf5dab008ff8e : SUCCESS in 2h 07m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/aafe736db5b446b999d36988db694f92 : FAILURE in 2h 29m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e483bc623b44f06ad1a45a60c868912 : SUCCESS in 1h 57m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/df36d291854349a5bff36ee9370ee3a1 : SUCCESS in 2h 37m 17s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/94adf56720024f54b644b352ee354ded : SUCCESS in 39m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7f00004011ae4e2bba1e1a74a1f81186 : SUCCESS in 1h 18m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c8585ae9c993426ebe343d93b7ecf4fb : FAILURE in 1h 41m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/835b4aecde884d56b505aa4f17b9d781 : SUCCESS in 44m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29d24f35aa654289ac509b23f139dad5 : SUCCESS in 2h 10m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63a609339318426a811657b0004fabd1 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e84d34cfe642152ed4198e6cd842389d734dfcc1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-03 12:13:53.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_06_2026_12_06_25\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"88171c9c4f4db0df89ee25a0701f5c4c2b0c2c7b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-03 13:03:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/afc5dcb437594e419dcbc3f13f83d0c8 : SUCCESS in 55m 40s\n- cinder-storpool-tempest-iscsi-multipath build/9d615d70d95e494aaa058dcce0cc77fb : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":4},{"id":"f1cf79e6594d97098142ea6cb9ee3324066982fd","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-03 15:01:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a5337162fae1bac26212dad8a9e196b052fa4bb","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-03 15:33:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9a5f55309ca9ef289490fc50fbf6a81910bef41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-04 03:52:13.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"962562efb7d7d934bfb97f5c0412cac7c355b0a8","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-04 04:16:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/fabca1db54b84388ae3aa3fb44f9c68e : FAILURE in 23m 14s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/988746","accounts_in_message":[],"_revision_number":5},{"id":"5e016d98f7f567e8ef5141e23cd9309f61a87537","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-04 04:30:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/1b8bb41853f54f6baa0111242e8cbc9a : SUCCESS in 36m 23s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/988746","accounts_in_message":[],"_revision_number":5},{"id":"0f5851593090bb8815b769cafeb8f6ac064a08e5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-04 04:34:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/73722b96e557474386f18542d73a8841 : SUCCESS in 39m 42s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/988746","accounts_in_message":[],"_revision_number":5},{"id":"cafef9f1fd1cb0357ff6d51433cfe35d1508b41f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-04 04:51:45.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibf0eb285aae37a456a367e7ec749543e7a743733 .","accounts_in_message":[],"_revision_number":5},{"id":"d754d41b5df371b172ed10e5a4280a2a055cbda3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-04 04:56:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/76edf82167ec4c728874d00d1f2bca29 : SUCCESS in 56m 03s\n- cinder-storpool-tempest-iscsi-multipath build/90897ae50b4c466082ddfc602fa867a0 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":5},{"id":"6a1a844bac37c1d6d5e1d76fb957df96b996e0e6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-04 04:58:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4349/.","accounts_in_message":[],"_revision_number":5},{"id":"dce3e047e6801d4211e6fce9062b2d03a8c51e27","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-04 04:59:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4173/.","accounts_in_message":[],"_revision_number":5},{"id":"c60f80a3d18f3fb7a890102fd95491326bb0012a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 06:04:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11a88c71ced741e1ad8aeeff6436a4e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7420a87cb2434ccbaa13a83902615ecd : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c4376b7cbc34d1ebe39aaa19f68f602 : SUCCESS in 7m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cf67d108af044cfa46a628911e737cc : SUCCESS in 8m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fdf027642ca746fdad67404324db4f4e : TIMED_OUT in 40m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572645a7470e4f089362738994604591 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/723755b8345a479e8a33db70bfe03d75 : SUCCESS in 31m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83d347c3389b484f8d7baa0febe32fd2 : SUCCESS in 49m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89e3d011fde146d28aadbc2542ac7055 : SUCCESS in 1h 41m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0e65f776b24474ab2388500ca480350 : SUCCESS in 1h 04m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7a738a9acca40179ab2193b7928924f : SUCCESS in 7m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bd395d6f6a374bf1951c3d6ea74702a4 : SUCCESS in 25m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/33d9b291cfcc4324b4df85fb0dac3682 : SUCCESS in 6m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e0fc9bb9ff0948c2b5938900155cb16f : SUCCESS in 5m 40s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/2149b229e0c548d8a81925d19361fa58 : FAILURE in 11m 07s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/1fb9eec28cda411abf03cf862a53aff0 : FAILURE in 13m 21s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1877057a1d3c4023b93d572629a0078f : SUCCESS in 7m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6071b406d0584fbb9c36757aed2fd039 : SUCCESS in 6m 27s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ede07408725b4e4d880aa5be6620db36 : SUCCESS in 6m 15s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c942ef3987cc48c99f62ad4b2dfc0dfc : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee4f8f6275764ea0bde617a2f1f9275c : SUCCESS in 4m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0cd687ec84ba443cbeffb4ac3a4a7e47 : SUCCESS in 1h 33m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d4bedee06578461b9a265d26e3bb110f : FAILURE in 1h 33m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/23334112c08d476ca2f23dee0fe61228 : FAILURE in 2h 06m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0d95ad59c7c74aba84269729a3c173ea : SUCCESS in 1h 46m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cc9a81a162434bb1b9b97501bad5a6ea : SUCCESS in 40m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a4d800c6a9e4e2d98e22803a1e1566a : SUCCESS in 1h 10m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e93deb2a06e48dd831024de29a2cfc0 : SUCCESS in 1h 41m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5af934118989499ab4f043607487dfb4 : SUCCESS in 46m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97a7a3f7ef05484eb62eb11780ab5d35 : SUCCESS in 54m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d03c5666022e4a818fc60de869b29d22 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":5},{"id":"73b0ed167f43722d2e36d6eeed6d6872f19484d7","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-04 06:36:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"40b9d4704bb8a5462b464c65f1197360ce0f436a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 07:07:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08b2516e95b4987a460a03b8d7fa104\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3c1fdb928a94655a25ea3e05ce56d19 : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/afc1e6a9ecf9479e869d3ae654fcbc3b : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/20f496edb2f34c3cb44267ebdc51a365 : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e9e648294f4a8066d98b0599190ff127d8e8d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:17:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d4104b9d0a74706a3f2828e3c3e3822\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced49a3cbccf47648a4aef8764face53 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/beff81ec58854aeaa27e9cb755a73766 : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f488b70b7e0a404d90b8b655dd5c5317 : SUCCESS in 9m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ad439c1921446a8a090572ac4280c1c : SUCCESS in 16m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2683b226534aaeabaa3bc285e07a52 : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b47958504b8b48c09ada3a51e7b61027 : SUCCESS in 54m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3659e41a2024ce4b976ddf9378341a6 : SUCCESS in 54m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82b647831a4e40679562e1994c06e8ee : SUCCESS in 1h 51m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48cee123e0c74f6c83cb93df8b85fc42 : FAILURE in 56m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7cccd31d8a24683bea729074db2dce6 : SUCCESS in 7m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ec6f3d0c7ab9401c8f181950891aff5e : SUCCESS in 30m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92e448075dee4ec48d359408901ed237 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59078812989f4890a296df73d42ab506 : SUCCESS in 4m 28s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/c96d55c8afe442259108ad5c23c7d8a3 : FAILURE in 15m 07s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/6f7c18979098431ab208e1c89598fcd3 : FAILURE in 11m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b059d0d4be0c4c8f9c93d052a06f8399 : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/12d1d0867ed04324837e9e0c844b97ee : SUCCESS in 5m 36s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cf0766edcbf343f7a394d34b478da90a : SUCCESS in 9m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5536b125ab2471b98498005b15f792f : SUCCESS in 29m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f980db1f4924c3892cb5b74cdfbe343 : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a32b4aae1c9c4d77b08107940ffe9499 : SUCCESS in 1h 30m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/099da08808bb476b85587eaa760dea6f : SUCCESS in 2h 35m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c30bc83417f248f9ab9513a4e53609b1 : SUCCESS in 1h 54m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/dca44d8e00a749a6a05d573e790fe5a9 : SUCCESS in 1h 38m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2618d8a46cf542039666df9774490061 : SUCCESS in 24m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2b5f4173c89d493f9e30b9ffcd4050c3 : SUCCESS in 1h 06m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/adcc532f105b4237899b94706f3f589e : SUCCESS in 1h 14m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1a8c583214224b56a21929fdb84dd6e6 : SUCCESS in 32m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bf87ba155304ac99f12453409286f2e : SUCCESS in 55m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93f49ad19ce2409c9932f9b02cd9af99 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c298b76b02e03d84b98fdacc2e1ed033da6b195d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-04 10:52:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/92579a7abb8949b78f6f6278066a23b9 : SUCCESS in 59m 29s\n- cinder-storpool-tempest-iscsi-multipath build/709527e764214b32b91f29cabddfc130 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":5},{"id":"67beb1a8276b916b4f4e10dbb6a08c8cac5c65e2","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-04 11:53:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2e0f302abc75de154c766c8b7787c43ede8cfccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:31:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1776a03f25c74ff1b07edd7a744d87a2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ae8469c072e48fdb70030f6ac88ee30 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e9fce02dfe8b46558c9b359a1aca465f : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/61d72476c2ab4e4c8ce97ed2266ee111 : SUCCESS in 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bdca550608ec719bbf8fa20bb7bda6b0d3a0eeac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:33:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80a7f5c33cc4c3c86696238aed6bcf3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9022409f2ff74c8191ab93e105becb4d : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec237cc7d0f740af9d277a8318d750d4 : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65a404b3bba8486eb88febc8318b6c37 : SUCCESS in 8m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e1c7c229cb9e44f48ea3bb84930f7942 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8156dc14df642e5821d66f1365c0a6b : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/44a45528bab9418c8870a0e1936eb9c2 : SUCCESS in 51m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f39bafe696534ff9a8380cf7aa42da48 : SUCCESS in 1h 11m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e2c2604069943659b741b96be57f5ad : SUCCESS in 1h 28m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/971ce0024f2e4ef59404e49e235dad71 : SUCCESS in 1h 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b836acbc091143648cbd4bdb77eec838 : SUCCESS in 10m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95b6dd49a5924fac8ae020908ebc9a43 : FAILURE in 23m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b81be277ae3940eea80e9d19d9f26ce6 : SUCCESS in 2m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/120153f394c24a3183faa4e9f753b876 : SUCCESS in 6m 39s (non-voting)\n- cinder-tox-py313-no-eventlet https://zuul.opendev.org/t/openstack/build/03b78eb3ecf64e2a91ce083caf7b7537 : FAILURE in 12m 59s (non-voting)\n- cinder-tox-py314-no-eventlet https://zuul.opendev.org/t/openstack/build/ee583b6b564541329401236a17dea521 : FAILURE in 12m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ffd1ba309f74519a5542253f17f2880 : SUCCESS in 7m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/98f26a7519594df69db4bd38ac3b0607 : SUCCESS in 6m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/550960f46869464b94f54cb91ee05185 : SUCCESS in 9m 33s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/58f6fef1d2564ea0adc906ff1e8ad30e : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a5048c4fb494eeb92a66abdf3aff9ac : SUCCESS in 3m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ad863f932104cbbba8d048e190df0a6 : SUCCESS in 1h 50m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/965359f60fbf48e8b2ca7d0a9c385506 : FAILURE in 2h 30m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/228b8eb061df471fbfdf961da5229891 : SUCCESS in 2h 01m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7ebc72a6afbe40b0ba04fbeaa15a532a : SUCCESS in 2h 10m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e642aeddff8549deb0fba1ec7c8d12f5 : SUCCESS in 44m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6537dd3d093c4637ae190e0067ae2d9b : SUCCESS in 1h 10m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b93b9c961923429087b1e48f285e7a9d : SUCCESS in 2h 05m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bd3f4f9dca143e29987a6a473d87684 : SUCCESS in 25m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc20f5dce2aa4a35be594b470f97e744 : SUCCESS in 1h 44m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f228c1b514e0490f9eaa6617a4e4ae8e : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":5},{"id":"91be2f84bc0a52fefd9cf392e1279949b4eccb6f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-04 14:54:40.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/04_06_2026_14_47_08\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"0452a59cd49441b843319949afa403422a392cb6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-04 15:14:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d467027274954a8ea6966199fc85f241 : SUCCESS in 55m 32s\n- cinder-storpool-tempest-iscsi-multipath build/1697b2a2899b418ebbb384b8b0121bbc : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":5},{"id":"6b8aee4bb855cd34a4630883c9a8a7aa236b4c20","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-09 13:28:26.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"41ae86166be9268b261947f350b133d84382ae96","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 13:01:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a6ebe39a89a7468c18ae9527fb3baa96d6f4c4d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-11 13:01:35.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/cinder/+/988747","accounts_in_message":[],"_revision_number":5},{"id":"600536b0585c1ae5ba5c8d363735bb672dccb967","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:59:57.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"604214fb8cc9c10597061149f81f268510db7e1c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:01:25.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":5,"current_revision":"8a1dafca919f3fb8d8c25b2bd87990118585b629","revisions":{"c301aef9e16472ba1eb2e59a9659a24189be38b0":{"kind":"REWORK","_number":1,"created":"2026-05-15 10:34:27.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/46/988746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988746/1"}}},"commit":{"parents":[{"commit":"d7515145399b45d23ff00e64b5ef9fc6ae55788e","subject":"Merge \"NetApp ONTAP: honor ONTAP 9.15.1+ default for LUN space-allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7515145399b45d23ff00e64b5ef9fc6ae55788e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 09:59:53.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 10:32:58.000000000","tz":330},"subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","message":"Add CI jobs to run unit tests with native threading (no-eventlet)\n\nThis adds unit test jobs for eventlet removal testing as part of the\nopenstack-wide goal to remove eventlet.\n\nThe added jobs set OS_CINDER_DISABLE_EVENTLET_PATCHING\u003d1 to disable\neventlet and use oslo.service threading instead.\nBoth jobs are non-voting to discover threading compatibility issues\nwithout blocking new patches.\n\nChange-Id: Ibf0eb285aae37a456a367e7ec749543e7a743733\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c301aef9e16472ba1eb2e59a9659a24189be38b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c301aef9e16472ba1eb2e59a9659a24189be38b0"}]},"branch":"refs/heads/master"},"f104218c5101033f93dd0864b7392b9cd292d26f":{"kind":"REWORK","_number":2,"created":"2026-05-27 13:06:47.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/46/988746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988746/2"}}},"commit":{"parents":[{"commit":"d7515145399b45d23ff00e64b5ef9fc6ae55788e","subject":"Merge \"NetApp ONTAP: honor ONTAP 9.15.1+ default for LUN space-allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7515145399b45d23ff00e64b5ef9fc6ae55788e"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 09:59:53.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-27 12:50:39.000000000","tz":330},"subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","message":"Add CI jobs to run unit tests with native threading (no-eventlet)\n\nThis adds unit test jobs for eventlet removal testing as part of the\nopenstack-wide goal to remove eventlet.\n\nThe added jobs set OS_CINDER_DISABLE_EVENTLET_PATCHING\u003d1 to disable\neventlet and use oslo.service threading instead.\nBoth jobs are non-voting to discover threading compatibility issues\nwithout blocking new patches.\n\nAlso skip unit tests that verify eventlet-specific behavior (tpool,\ngreenpool, eventlet.Semaphore) when running in threading mode.\n\nAlso skips eventlet-specific unit tests when running in threading mode,\nthese tests verify eventlet impl details not used with native threads.\n\nChange-Id: Ibf0eb285aae37a456a367e7ec749543e7a743733\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f104218c5101033f93dd0864b7392b9cd292d26f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f104218c5101033f93dd0864b7392b9cd292d26f"}]},"branch":"refs/heads/master"},"1d23f8b50f48fafbd5c92837b1162fa644e4dbbe":{"kind":"REWORK","_number":3,"created":"2026-05-29 15:56:42.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/46/988746/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988746/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988746/3"}}},"commit":{"parents":[{"commit":"2ef4fadbca5ed10390926e56bba309cf34b4d4a0","subject":"Merge \"Fix cascade delete for non-admin with deleted cross-project snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ef4fadbca5ed10390926e56bba309cf34b4d4a0"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 09:59:53.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 15:52:36.000000000","tz":330},"subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","message":"Add CI jobs to run unit tests with native threading (no-eventlet)\n\nThis adds unit test jobs for eventlet removal testing as part of the\nopenstack-wide goal to remove eventlet.\n\nThe added jobs set OS_CINDER_DISABLE_EVENTLET_PATCHING\u003d1 to disable\neventlet and use oslo.service threading instead.\nBoth jobs are non-voting to discover threading compatibility issues\nwithout blocking new patches.\n\nAlso skip unit tests that verify eventlet-specific behavior (tpool,\ngreenpool, eventlet.Semaphore) when running in threading mode.\n\nAlso skips eventlet-specific unit tests when running in threading mode,\nthese tests verify eventlet impl details not used with native threads.\n\nChange-Id: Ibf0eb285aae37a456a367e7ec749543e7a743733\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d23f8b50f48fafbd5c92837b1162fa644e4dbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d23f8b50f48fafbd5c92837b1162fa644e4dbbe"}]},"branch":"refs/heads/master"},"c4546494133a0afee5f38dac766a2c7cd9babe66":{"kind":"REWORK","_number":4,"created":"2026-06-03 04:52:05.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/46/988746/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988746/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988746/4"}}},"commit":{"parents":[{"commit":"a0fa5b2fe3179e8374c41000fdd354078ba49cc6","subject":"Merge \"Remove mypy-wrap.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0fa5b2fe3179e8374c41000fdd354078ba49cc6"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 09:59:53.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-06-03 04:46:37.000000000","tz":330},"subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","message":"Add CI jobs to run unit tests with native threading (no-eventlet)\n\nThis adds unit test jobs for eventlet removal testing as part of the\nopenstack-wide goal to remove eventlet.\n\nThe added jobs set OS_CINDER_DISABLE_EVENTLET_PATCHING\u003d1 to disable\neventlet and use oslo.service threading instead.\nBoth jobs are non-voting to discover threading compatibility issues\nwithout blocking new patches.\n\nAlso skip unit tests that verify eventlet-specific behavior (tpool,\ngreenpool, eventlet.Semaphore) when running in threading mode.\n\nAlso skips eventlet-specific unit tests when running in threading mode,\nthese tests verify eventlet impl details not used with native threads.\n\nChange-Id: Ibf0eb285aae37a456a367e7ec749543e7a743733\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4546494133a0afee5f38dac766a2c7cd9babe66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4546494133a0afee5f38dac766a2c7cd9babe66"}]},"branch":"refs/heads/master"},"8a1dafca919f3fb8d8c25b2bd87990118585b629":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-04 03:52:13.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/46/988746/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/46/988746/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/46/988746/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/46/988746/5"}}},"commit":{"parents":[{"commit":"a0fa5b2fe3179e8374c41000fdd354078ba49cc6","subject":"Merge \"Remove mypy-wrap.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a0fa5b2fe3179e8374c41000fdd354078ba49cc6"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-15 09:59:53.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-06-04 03:47:35.000000000","tz":330},"subject":"Add CI jobs to run unit tests with native threading (no-eventlet)","message":"Add CI jobs to run unit tests with native threading (no-eventlet)\n\nThis adds unit test jobs for eventlet removal testing as part of the\nopenstack-wide goal to remove eventlet.\n\nThe added jobs set OS_CINDER_DISABLE_EVENTLET_PATCHING\u003d1 to disable\neventlet and use oslo.service threading instead.\nBoth jobs are non-voting to discover threading compatibility issues\nwithout blocking new patches.\n\nAlso skip unit tests that verify eventlet-specific behavior (tpool,\ngreenpool, eventlet.Semaphore) when running in threading mode as these\ntests verify eventlet impl details not used with native threads.\n\nChange-Id: Ibf0eb285aae37a456a367e7ec749543e7a743733\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a1dafca919f3fb8d8c25b2bd87990118585b629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8a1dafca919f3fb8d8c25b2bd87990118585b629"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
