)]}'
{".zuul.yaml":[{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"4e8ee6b65589653c95f968f717f563d648352298","unresolved":true,"context_lines":[{"line_number":33,"context_line":"        # Python 3.14 testing is optional but encouraged in Hibiscus"},{"line_number":34,"context_line":"        - openstack-tox-functional-py314:"},{"line_number":35,"context_line":"            voting: false"},{"line_number":36,"context_line":"            irrelevant-files: *functional-irrelevant-files"},{"line_number":37,"context_line":"        - cinder-rally-task:"},{"line_number":38,"context_line":"            voting: false"},{"line_number":39,"context_line":"            irrelevant-files: *functional-irrelevant-files"}],"source_content_type":"text/x-yaml","patch_set":8,"id":"a890e89b_cf577d65","line":36,"updated":"2026-06-11 13:02:41.000000000","message":"to have similarity with other jobs","commit_id":"8f3682471be7870be660204410d63d17b46e905f"},{"author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"change_message_id":"bde9fbd38cc471a4892f011168cf8e56f59f9321","unresolved":true,"context_lines":[{"line_number":33,"context_line":"            voting: false"},{"line_number":34,"context_line":"            irrelevant-files: *functional-irrelevant-files"},{"line_number":35,"context_line":"        # Python 3.14 testing is optional but encouraged in Hibiscus"},{"line_number":36,"context_line":"        - openstack-tox-functional-py314:"},{"line_number":37,"context_line":"            voting: false"},{"line_number":38,"context_line":"            irrelevant-files: *functional-irrelevant-files"},{"line_number":39,"context_line":"        - cinder-rally-task:"}],"source_content_type":"text/x-yaml","patch_set":12,"id":"dd5c1911_65b6977d","line":36,"updated":"2026-06-28 15:12:31.000000000","message":"any plans to add threading job for py314?","commit_id":"798b1a1d427bd415cbe24be2a201ea6e38ca489d"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"fbd176f1df13691167dfc132eec81ec097afee91","unresolved":true,"context_lines":[{"line_number":33,"context_line":"            voting: false"},{"line_number":34,"context_line":"            irrelevant-files: *functional-irrelevant-files"},{"line_number":35,"context_line":"        # Python 3.14 testing is optional but encouraged in Hibiscus"},{"line_number":36,"context_line":"        - openstack-tox-functional-py314:"},{"line_number":37,"context_line":"            voting: false"},{"line_number":38,"context_line":"            irrelevant-files: *functional-irrelevant-files"},{"line_number":39,"context_line":"        - cinder-rally-task:"}],"source_content_type":"text/x-yaml","patch_set":12,"id":"5e0a3ee7_be19298c","line":36,"in_reply_to":"dd5c1911_65b6977d","updated":"2026-06-30 07:21:42.000000000","message":"yes, we will later, right now default (non threading) 3.14  is also non-voting, \nenabling threading job for 3.14 will just increase our CI time.\n\nI guess first we will work on fixing test with existing support of 3.12, 3.13, then will enable 3.14","commit_id":"798b1a1d427bd415cbe24be2a201ea6e38ca489d"},{"author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"change_message_id":"06540b17bfa5f3990e29d5bd124e13adaf1c9bcd","unresolved":true,"context_lines":[{"line_number":201,"context_line":"        \u0027SYSTEMD_ENV_VARS[\"c-bak\"]\u0027: OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue"},{"line_number":202,"context_line":""},{"line_number":203,"context_line":"- job:"},{"line_number":204,"context_line":"    name: cinder-tox-py313-threading"},{"line_number":205,"context_line":"    parent: openstack-tox-py313"},{"line_number":206,"context_line":"    description: |"},{"line_number":207,"context_line":"      Run Cinder unit tests with native threading instead of eventlet."}],"source_content_type":"text/x-yaml","patch_set":12,"id":"7ad915c2_1f75ee5f","line":204,"updated":"2026-06-23 14:45:49.000000000","message":"We probably don\u0027t need this since we added https://review.opendev.org/c/openstack/cinder/+/982144 ?","commit_id":"798b1a1d427bd415cbe24be2a201ea6e38ca489d"},{"author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"change_message_id":"34e1d3adea9ef6110bb27332b2f3006fca500709","unresolved":true,"context_lines":[{"line_number":201,"context_line":"        \u0027SYSTEMD_ENV_VARS[\"c-bak\"]\u0027: OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue"},{"line_number":202,"context_line":""},{"line_number":203,"context_line":"- job:"},{"line_number":204,"context_line":"    name: cinder-tox-py313-threading"},{"line_number":205,"context_line":"    parent: openstack-tox-py313"},{"line_number":206,"context_line":"    description: |"},{"line_number":207,"context_line":"      Run Cinder unit tests with native threading instead of eventlet."}],"source_content_type":"text/x-yaml","patch_set":12,"id":"ec8c03b9_8a788ccc","line":204,"in_reply_to":"74d7fe09_0a9ed017","updated":"2026-07-23 14:51:21.000000000","message":"I understand but I don\u0027t think we want to add both of them right now. Maybe after we get one working.","commit_id":"798b1a1d427bd415cbe24be2a201ea6e38ca489d"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"fbd176f1df13691167dfc132eec81ec097afee91","unresolved":true,"context_lines":[{"line_number":201,"context_line":"        \u0027SYSTEMD_ENV_VARS[\"c-bak\"]\u0027: OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue"},{"line_number":202,"context_line":""},{"line_number":203,"context_line":"- job:"},{"line_number":204,"context_line":"    name: cinder-tox-py313-threading"},{"line_number":205,"context_line":"    parent: openstack-tox-py313"},{"line_number":206,"context_line":"    description: |"},{"line_number":207,"context_line":"      Run Cinder unit tests with native threading instead of eventlet."}],"source_content_type":"text/x-yaml","patch_set":12,"id":"74d7fe09_0a9ed017","line":204,"in_reply_to":"7ad915c2_1f75ee5f","updated":"2026-06-30 07:21:42.000000000","message":"thats adding for 3.12 `openstack-tox-py312`, this is 3.13","commit_id":"798b1a1d427bd415cbe24be2a201ea6e38ca489d"}],"/COMMIT_MSG":[{"author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"change_message_id":"24922f50d029e40295ec2851aeb3f63f38c6d4f4","unresolved":true,"context_lines":[{"line_number":9,"context_line":"Add non-voting unit and functional test jobs that set"},{"line_number":10,"context_line":"OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue to test threading mode."},{"line_number":11,"context_line":""},{"line_number":12,"context_line":"Also rename cinder-without-eventlet to cinder-tox-py312-thread"},{"line_number":13,"context_line":"Also add monkey_patch call in functional test __init__.py to ensure"},{"line_number":14,"context_line":"proper initialization in threading mode."},{"line_number":15,"context_line":""}],"source_content_type":"text/x-gerrit-commit-message","patch_set":17,"id":"91377770_db346b94","line":12,"updated":"2026-08-16 14:28:04.000000000","message":"cinder-tox-py312-thread to cinder-tox-py312-threading","commit_id":"4cfbee1b913f765c9529e19a9a2bcf0b14ab2806"}],"/PATCHSET_LEVEL":[{"author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"change_message_id":"696fb4c4a0891a03526380007cb3adc3dbc10647","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":1,"id":"35372245_3182e009","updated":"2026-05-15 14:25:38.000000000","message":"cinder-tox-functional-py313-no-eventlet appears to not work","commit_id":"fcb9efa7ff67564a7c7c5eb1960a0287f7a9df75"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"806d3c2b8dd3181e59b2930a892e4bca13d140a8","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":1,"id":"8db177b8_c64e28b9","in_reply_to":"35372245_3182e009","updated":"2026-05-28 06:04:08.000000000","message":"fixed now\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_af5/openstack/af5ba3f541fa4640b92515088e66f3fc/testr_results.html\n\nfunctional tests are importing monkey_patch too.\n\nI\u0027ll make the job no-voting in next PS, after reabsing with https://review.opendev.org/c/openstack/cinder/+/981107","commit_id":"fcb9efa7ff67564a7c7c5eb1960a0287f7a9df75"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"806d3c2b8dd3181e59b2930a892e4bca13d140a8","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":2,"id":"15bd0419_30b04739","updated":"2026-05-28 06:04:08.000000000","message":"recheck\n\ncinder-mypy job is unblocked now","commit_id":"7885e3b7dde763f614716f5be3cc3c8c75b38162"},{"author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"change_message_id":"67c19e5c444713a36c273f3e7ea5803a293c2e8f","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":10,"id":"82fff364_7be053ae","updated":"2026-06-16 22:10:48.000000000","message":"tox~threading looks good, tox-function-threading is running but 100% fail rate - is that expected for this change?","commit_id":"0e1ebdb8d713c567db2cce8a3e60ee2282d669fc"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"115f18168428025a6a875d6be40af5951ee6d965","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":10,"id":"f8e5b6cf_4f04fd7b","in_reply_to":"82fff364_7be053ae","updated":"2026-06-17 07:23:00.000000000","message":"yes, functional tests fails are expected, these patches are to only to add the CI job, as non-voting to start discovering threading compatibility issues.\nwe need to work on fixing these in follow up patches, so eventually promote these jobs to voting once its stable.","commit_id":"0e1ebdb8d713c567db2cce8a3e60ee2282d669fc"},{"author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"change_message_id":"693d5e1670600196113cdfb60bcab4668ee000d5","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"705fc48c_291c12df","updated":"2026-06-30 10:23:31.000000000","message":"recheck \nunreated tests failed for py3.13 job\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_7ae/openstack/7ae1146dba6c46d69cf2cfc24964c4fa/testr_results.html","commit_id":"3c77f8457a69442aba1377e3e3335b28de1881b9"},{"author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"change_message_id":"24922f50d029e40295ec2851aeb3f63f38c6d4f4","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"c915ba53_0f2bacc1","updated":"2026-08-16 14:28:04.000000000","message":"Nit: typo cinder-tox-py312-thread to cinder-tox-py312-threading","commit_id":"4cfbee1b913f765c9529e19a9a2bcf0b14ab2806"},{"author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"change_message_id":"2dc538c7a781334e6202a40e5c2fc25ce0459a8d","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"7ec40297_899da753","updated":"2026-08-12 13:25:11.000000000","message":"Overall, LGTM.\n\nLooks like cinder-tox-functional-py313-threading needs some work, though it looks like the issues are in the functional test harness itself.  But it\u0027s non-voting and we\u0027ll have to get it running eventually, so I think it\u0027s ok to merge.","commit_id":"4cfbee1b913f765c9529e19a9a2bcf0b14ab2806"}]}
