)]}'
{"id":"openstack%2Fcinder~989051","triplet_id":"openstack%2Fcinder~master~Ifd7d943f3d28fd072829bec351fd3dba2c16d32f","project":"openstack/cinder","branch":"master","topic":"bug/2153099","attention_set":{"37798":{"account":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"last_update":"2026-05-29 15:29:59.000000000","reason":"Someone else replied on the change"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-05-29 03:41:34.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"38702":{"account":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"last_update":"2026-06-19 20:02:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ifd7d943f3d28fd072829bec351fd3dba2c16d32f","subject":"Always delete silent objects left behind when RBD parents objects are deleted.","status":"NEW","created":"2026-05-19 03:34:02.000000000","updated":"2026-06-19 20:05:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f1c5f7cff0a37ed4bbd723e844e103dc81d490a8","_number":989051,"virtual_id_number":989051,"owner":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-29 16:29:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 04:26:33.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-19 04:32:33.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-19 04:39:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-19 05:00:05.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-05-19 05:25:44.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-19 05:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-19 05:57:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-19 05:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 07:21:30.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-22 12:06:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-22 16:29:14.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-05-29 21:47:07.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-19 19:49:37.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"b9d8b9bf3cb2ef464973452897e5704221b6b93f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-19 03:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38f7f305042d29dd11cc46b458d4e57ae036eb9","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-19 04:26:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-fc-driver build/f75769e42f204b0b8d61f2d2bfee4be1 : FAILURE in 31m 17s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/989051","accounts_in_message":[],"_revision_number":1},{"id":"76919984e5f06be89054230d3b63536ac5d4a9b7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-19 04:32:33.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifd7d943f3d28fd072829bec351fd3dba2c16d32f .","accounts_in_message":[],"_revision_number":1},{"id":"32e7f7ba205517bc710b5824cf429f2b15eefdd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-19 04:39:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/b3944344482847008d35229668e1fcb1 : FAILURE in 29m 19s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051","accounts_in_message":[],"_revision_number":1},{"id":"1597912fabd46fd459ea784507ec24a2ec4af409","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-19 05:00:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/550c18597b6d4311a37020818007b978 : FAILURE in 37m 49s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/989051","accounts_in_message":[],"_revision_number":1},{"id":"05485dc53fbd7d653c229bee70859eccf1fc91ea","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-19 05:25:44.000000000","message":"Patch Set 1:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/989051 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"06d82bdd397a57070b2f7b1a0a1bce2a4502f904","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 05:28:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf5e9351c1fc47bbae60e0447c301b77\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/20480cbc7fe84bfa8a6c00003ea6dd7d : FAILURE in 12m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f758cfd8a92d45b6b4c6c8170ff2832d : FAILURE in 10m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95b2f243ebe94e6f905d71efb72d80ee : FAILURE in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5351222499c5f710d58ed2bfbc8303fbeb38e480","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-19 05:57:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5b9fc0a52f394e938ba14a8bbfdbd278 : SUCCESS in 49m 08s\n- cinder-storpool-tempest-iscsi-multipath build/f6ea564b0388410dbcfffa484a1d4800 : FAILURE in 53m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fc0d337bfff4f7a22247b6fb3c69ee8a8b22771a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 05:59:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/421986a1490c4988b7b37d634e2f8164\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9cbb9623754fde8b990c6c7f2206b3 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7765dd5006bf4741acc8ea0ef5a21fad : FAILURE in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3c8d4445d754e7eaa5d316c3ee5860b : FAILURE in 7m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c91697be8e1453fbe793ca9f236f0bf : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f5a37e659d4e1ab1047ecc3da5cfa0 : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d62f538083bd4c41840ff94395a58181 : SUCCESS in 30m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3271429f92424792816e4f152ddaddcc : SUCCESS in 49m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3bf0db99d0e4e98b345954fabd08313 : SUCCESS in 1h 39m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/195c0a6d24bc4c90909a275ea35a1e49 : SUCCESS in 48m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d94e5fa45c0645f1b1f277719c8d1f56 : SUCCESS in 9m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d89b1ddeeee4a8ab9ff822c4cc3fcd2 : FAILURE in 18m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6c2e9d5d2d9a47e29dbbca25da64d99b : SUCCESS in 4m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ba457886290411a95e4fa1384e78591 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0a440b5ed1ef4c6b8c5ed4174652c2cd : SUCCESS in 5m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5359ac7dd5bf4112b48e2a87d89194ba : SUCCESS in 5m 12s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4247cc42f78549cba141c06adc876300 : SUCCESS in 11m 35s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60ccf3479ab440568392122d8b646159 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c5bed1d14b542bab795a3cdd3c2889d : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ee9c44405124c549e590f123a97f02b : SUCCESS in 1h 30m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5df4ed636aa441a0b7cdfadec3c7cee9 : FAILURE in 2h 09m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e8e7e553201d4ebe8f5ddb457203de36 : SUCCESS in 1h 53m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/18648ec6393946b48dc32b20a98090f3 : RETRY_LIMIT in 2m 34s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2df08c3ba1b4461dadcbbcdf9ff5e03d : SUCCESS in 34m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/319aa55c4ee64b97b0f14dd4654634fe : SUCCESS in 1h 03m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/277c2a19acc241268eea7e2d7520c1a3 : SUCCESS in 1h 39m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3532a99b22bd4d6cb4a75b46123f616f : SUCCESS in 27m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34805d766d5a4d749ea45536a68c4716 : SUCCESS in 44m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df2af6499a3d4a128e579297b283984d : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6253e97d4c5f154db82400f5f03d3bc1ef60cb6f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 07:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4187/.","accounts_in_message":[],"_revision_number":1},{"id":"daf120d1d97be69c9b7301f074589bc14b749a66","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 07:27:47.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4011/.","accounts_in_message":[],"_revision_number":1},{"id":"6a14fef01738eb3c96d6fb57dfcb9d66a1c910c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-19 19:52:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c45889810e3c8807e4b1e91e26bdf7a689e268d4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-19 19:53:05.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update opendev/openstack/cinder","accounts_in_message":[],"_revision_number":2},{"id":"7783957cf7a4aac94feb22c708092005754f4df3","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-19 20:17:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/36d629ee79ec488e94ab98653913890d : FAILURE in 24m 41s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/989051","accounts_in_message":[],"_revision_number":2},{"id":"daa311bb8968854d514828be5807c153ac6fca60","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-19 20:21:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/dfba9cb2bf3f447686f43115681e78a1 : FAILURE in 28m 26s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051","accounts_in_message":[],"_revision_number":2},{"id":"56c83e5a8def620dbbc99f7fd22b9dd5a516f038","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-19 20:27:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-fc-driver build/80498ab0452e4989bf76660d3bacdd7b : FAILURE in 34m 58s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/989051","accounts_in_message":[],"_revision_number":2},{"id":"669aa6f7be8ebccc8f019e9caaa383637cc2e67d","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-19 20:40:08.000000000","message":"Patch Set 2:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/989051 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a23ac3ddaba29e17eb37a73d1747ae5421b3e7ee","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 21:04:19.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4016/.","accounts_in_message":[],"_revision_number":2},{"id":"6588674d32c7f85e0dda773ca27e698a632291df","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-19 21:04:19.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4192/.","accounts_in_message":[],"_revision_number":2},{"id":"ca1603a03d7d582ea225872e9ef1e70fdf4641a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 21:55:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1082d9e27d0244e08f158b6a8f36d59b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943f2ee41d01416ea0f4c2428989ab55 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baa915e8ae5440ea85e8e4f030bf9050 : SUCCESS in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e18235ac318a49e5a6fbddcd53641abd : SUCCESS in 11m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe2de005170940d98df56c08ad7bdfd9 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248e67ba3d854fad8d4b4b40fa171823 : SUCCESS in 16m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/d72a8af7a9194fabb2c34bd19abbba4d : SUCCESS in 1h 15m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b506bac00c664b2c9b6f121fc7ab7720 : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a4369523c1342eeb7ccb1f9455108a5 : SUCCESS in 1h 27m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77209e9f9e594dceb907ffdecc20e5d3 : SUCCESS in 1h 32m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/331f9eecaf8b45858b3afc32fc90528a : SUCCESS in 10m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6049237c56bc4509872f1808fb2736eb : TIMED_OUT in 41m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ee22a26b731d4e5388167190a013c036 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5daf26020e0464a892f0f5938e98b5b : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d135bd29e9524af294a0a0e8ba496496 : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/36adff13eb034ea392ab91eff8fff7eb : SUCCESS in 7m 49s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8414fa5d58ab48ac9055b9398968948c : SUCCESS in 9m 20s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aa60ee5bea3e43f4a6752024b21c8345 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3dae31b636db46c09e120b5f41f230d4 : SUCCESS in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/feab3dd838954f0a81d07e63e2438597 : SUCCESS in 1h 01m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0606a14c8edf436ca7c46adae5136416 : FAILURE in 1h 41m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e30d96a57144959a0435594f375e74c : FAILURE in 1h 42m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e1590d234816479e8d80e9b3155c5ef2 : FAILURE in 1h 49m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e48b6497a2494ebbac2f799fc28cffc2 : SUCCESS in 20m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d2120bda64694cbe81bffd462230bec2 : SUCCESS in 1h 25m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0115bd415b1d4d1d91a09f05f62a257b : SUCCESS in 1h 46m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/20c098a96642479b8cdd6a78fceafe86 : SUCCESS in 37m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba0dc027b7464ce5ad52cbe2cb383a54 : FAILURE in 1h 40m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/560bc9e100494f588dfecd49d145eaca : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e9c8c7d365f48b05af3497b3a4c767d3a24f9ed7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 23:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb8a27aebc54592a78e5e71556830c2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3fbcf295a69a4b6a8d57d33fd01fdf21 : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/adb5edeb207b46158743b3321bdffcc2 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3a05d60c6d5d43758d1a8e12ffaf3194 : SUCCESS in 32m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb5285ff03590aa1352686edd3474963429cd28e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-22 01:18:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fd59ef7b052a07951504aa651d89c5eadbe3f3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 02:04:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcfeb59adbb046bca29d04cdae77b5b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f63cb2362dde472d9ca731738ae4c9c5 : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f855feb5c7048938a8e5006814d3cfa : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/08829da8fec34be4b0a0736e249f5516 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc6df634a55d5b616bd0a70e0b799d6de4c75ea5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 02:17:57.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifd7d943f3d28fd072829bec351fd3dba2c16d32f .","accounts_in_message":[],"_revision_number":3},{"id":"d677ac23afe7413965adec5b8c2ff53c7c48e41f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-22 02:25:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/7d0cfbdbd6ab406fb4032ea90bf0b953 : RETRY_LIMIT in 24m 13s\n- cinder-storpool-tempest-iscsi-multipath build/ae7268f28d994a508255ecd84c7b6045 : RETRY_LIMIT in 24m 12s","accounts_in_message":[],"_revision_number":3},{"id":"f4a5d2ba76f64c955db992bdc2c8ddd2c4e14e58","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-22 02:46:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/69356220523d43869ebda2deb2501e6a : FAILURE in 29m 01s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051","accounts_in_message":[],"_revision_number":3},{"id":"ffa6417924565793a0c8c72487c91035f1c00f1f","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-22 04:08:44.000000000","message":"Patch Set 3:\n\nBuild Succeeded: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/989051 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"05611201fb6c8e2b28f22ca367abbdaf27689940","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-22 04:42:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4043/.","accounts_in_message":[],"_revision_number":3},{"id":"f59feb204eb0e83521a927fc9c7924d750148997","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-22 04:49:19.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4219/.","accounts_in_message":[],"_revision_number":3},{"id":"f6317c4e7514af24e12859bdf8b13b833348725d","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-22 05:32:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver build/b9c640894d2f4a86a299b31616055452 : FAILURE in 38m 34s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/989051","accounts_in_message":[],"_revision_number":3},{"id":"4d516d3f8cd34950382ac8269488eb2823d337b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 12:06:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44e6c2174ce40f68ca27b3c1fa73d78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0561a47c5f234f33b7884b5b185384cc : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/612a97809455417880654d1ff89934d6 : SUCCESS in 8m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0f94ea0964747189838d9d0d93cd0d1 : SUCCESS in 10m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a268fa469a04a95bac895e3cf6543bc : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c4e862fbade4abab93535e6700edab6 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/71813d8fc7f44524a0395b7167b2e606 : SUCCESS in 1h 10m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8727eb8311e44cab140ffd438540a35 : SUCCESS in 1h 09m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0bb3b9779546426b9c1f72d0bf2038c5 : SUCCESS in 1h 55m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b109a42d21e4d19b80e1ee1e98f1d42 : SUCCESS in 1h 04m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72b5e94f3ee48f888f7fcc2ecec011a : SUCCESS in 5m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a604964b6f8d46ddb187a449515c54f3 : SUCCESS in 29m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/805d359b084f4d8eb684004cda27e22e : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04b5a357aabd4cd699c877f73346c487 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6092d3d47bd8496285af5bb63f04ac35 : SUCCESS in 5m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7fded36699924c43a0ea7fc0e896e07d : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ea99675f2db54662bec00fea48230625 : SUCCESS in 9m 00s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a7b626514feb46209e8f83c35ddacec2 : SUCCESS in 40m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6aa0a8033af84fdc97b752ba6653b1a1 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/021421dd673d4ce293a9bf6b3d861dc5 : SUCCESS in 1h 47m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b222c3644c3443798eeeef034f222618 : FAILURE in 2h 54m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d220ead123e24f549192f6b721f389c8 : SUCCESS in 2h 03m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/151501529c9c468fbe6605b8100453fc : SUCCESS in 1h 32m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2d09dc6bea714cab940418e6a58820e5 : SUCCESS in 36m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8bf733d456114046a1dca1f4ea0a36fc : SUCCESS in 1h 07m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b0914613e744611a0c116949f32fbb7 : SUCCESS in 1h 28m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7368fe6264144bc08681bfb701487c51 : SUCCESS in 26m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04a445cfb9084449b692024799005814 : SUCCESS in 2h 01m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/096c194472704fae8e1723da08383737 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":3},{"id":"38e6046a813b730d4f2bc2335cdbaea7bdf26fd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-22 12:16:22.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_05_2026_12_09_43\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"0e87fb20d02e268d7c6f8345787babd71aad2858","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-22 15:34:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/6508fe39d04b431888d7751a71509e26 : FAILURE in 49m 10s\n- cinder-storpool-tempest-iscsi-multipath build/19833e427b48432991e03ce30746b9fe : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":3},{"id":"9df98429384ab839f2ff265d135ef66e37be11d6","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-22 16:29:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ec5bb855fbd586e9bf55b3261b755cbad6bf921","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-27 01:06:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"661e331ac406df88d74eee035f0f944bb657fb02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 03:41:34.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ec9b171db3efaefc38d4d7c94fc886304a0fbabc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 04:16:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/4eda53e135494f539597757ad783633e : RETRY_LIMIT in 18m 54s\n- cinder-storpool-tempest-iscsi-multipath build/05863777fbdd4115b9cca0f6840cdbcc : RETRY_LIMIT in 18m 49s","accounts_in_message":[],"_revision_number":4},{"id":"784f0c146d9add7916bd31c23c9371bb6833462a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 04:18:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver build/ecddbd79eaa546ccb89a8309853f0b9f : SUCCESS in 35m 56s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/989051","accounts_in_message":[],"_revision_number":4},{"id":"7e1a79c19bc13bab27ffe1c75f9d2e6c65043243","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 04:28:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/6e4ce3db0f904dd2a245b271a3b3513d : SUCCESS in 46m 06s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/989051","accounts_in_message":[],"_revision_number":4},{"id":"ca155a62aec918b678e2ca6a45243ac376c02487","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 04:37:01.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifd7d943f3d28fd072829bec351fd3dba2c16d32f .","accounts_in_message":[],"_revision_number":4},{"id":"a45f2f54a3d8cede19ee5e63056c830dfebd9549","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 05:10:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4278/.","accounts_in_message":[],"_revision_number":4},{"id":"0bd1d90fd7c026ebaacf61cbe0ee7ba473614857","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 05:12:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/bd5e9c82659f4afc9d43e9251a037999 : SUCCESS in 40m 22s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051","accounts_in_message":[],"_revision_number":4},{"id":"6a577a3b46626f076bfc17acb96e64f8afdd5062","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 05:15:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4102/.","accounts_in_message":[],"_revision_number":4},{"id":"5298901dd07cf0b886f3ae826d6579c09e0574f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 07:08:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/170318b4e3644d5e804289c27df6ef81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc35b5801b54e0083b4f07f59e19a99 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/602226befbdc4f6cabc4cc6ea08ea265 : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9f7c7762dfa48818b4c19290a497b17 : SUCCESS in 6m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d2346ba2b61461ba0a306081bacb1c7 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ea46594ead41779394acd0388a9f10 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/82d3d7d7b9304666bd4c80d38a314e35 : SUCCESS in 1h 05m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3cfb58fb421c468ab7adb3cd870cc581 : SUCCESS in 36m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f1a53bb5b374f73b158120594e53f17 : SUCCESS in 1h 29m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9fef11f085841a3a9e8ee5988abbb71 : SUCCESS in 1h 08m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ac783b78a31459899242101e6ea9742 : SUCCESS in 11m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/168475217e4d4659941a976b275fb13b : SUCCESS in 22m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3a2fb2e9c9b845ee804d650bdc8bb564 : SUCCESS in 2m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b53254b1e4b4456da3b30480dca2ffe3 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0c410d0c21d43a0bffcc959f45ea5dd : SUCCESS in 5m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/452bb088931f474e9067fb45a1c6367c : SUCCESS in 7m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/535185fe821e41bf9830fae7b4cb3301 : SUCCESS in 8m 02s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8ad347cdefab4d2098d03d3379201f02 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f830e3d60e4a47d69c84c4742e50ffec : SUCCESS in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65d27fd345d748aba1d654340f5eadde : SUCCESS in 1h 09m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/00cd5b0eb449491aac61bf28b99b35bf : TIMED_OUT Host unreachable in 3h 20m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8c90eae40dcb4aef940cf643ab4f6dbc : FAILURE in 1h 40m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1fe4fdb3332b43adabeb7c6190c87452 : SUCCESS in 1h 49m 42s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0a133a9de2e84299a617140844e489ce : SUCCESS in 38m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/063a29d254ef4efab02db31c5820c6ef : SUCCESS in 41m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/40ba7d2535a4453f84962e1dcba07a85 : SUCCESS in 1h 45m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/246ebffb3a3c43cca60dc0fa5e4d1f1f : SUCCESS in 45m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eb98964b2e524d53bf2f2a1ab361948a : SUCCESS in 1h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd111f58b1404a3391da6e4d7d277e2a : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":4},{"id":"21587beb96a56175ea9ed0a1e130670444877f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 08:42:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd78e2f28079401cafea94053c001748\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/768962c67c67425f8a4ab004a092111f : SUCCESS in 24m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0028be8ef724be78bd7f51cc73b2312 : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98a495c9081e41aaa8b49a871d396fca : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3ae76840f58875427dbf5d89341140ece6f8f161","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"date":"2026-05-29 14:16:41.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f5a303da08bc43c6964502995152292bef71bbed","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-29 15:01:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/21f8d358d486466db604b5b98ce0a002 : RETRY_LIMIT in 17m 41s\n- cinder-storpool-tempest-iscsi-multipath build/2efb3ccd6b1d4042bdf6e7121e3ca45b : RETRY_LIMIT in 18m 59s","accounts_in_message":[],"_revision_number":5},{"id":"683d475952ce3073abaf1a0c99d3f95754a11d34","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 15:14:03.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifd7d943f3d28fd072829bec351fd3dba2c16d32f .","accounts_in_message":[],"_revision_number":5},{"id":"4bbd4caf2ccd9bad86baf53fa356155b400ba37a","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-29 15:29:59.000000000","message":"Patch Set 5:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/989051 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"99eb8526451e9892ff6d0a1168047534039a086f","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-29 16:05:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/af798adcea624bcdb7d665ed722ca12a : SUCCESS in 46m 30s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/989051","accounts_in_message":[],"_revision_number":5},{"id":"75bd70dca63f85fcdeaeed4d30b8c2cd748080ff","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-29 16:19:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-fc-driver build/158f9f2d4f1c49869c21ce294e881ff7 : SUCCESS in 36m 42s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/989051","accounts_in_message":[],"_revision_number":5},{"id":"bf12c3aeb1b008e68f6338197806ed9b96cde651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 16:29:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df7af2ccac4e43c8a53bb994b8f5fcef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab053dcfb21476899970e3f43ce4e2e : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63132665fcd04b8699ba8083acd3bf3e : SUCCESS in 8m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4983a722ec614735acf8ab9b58e788ae : SUCCESS in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e75e7e8ce26c47189f16359fe494690a : SUCCESS in 11m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8268a2ff1ed54213bbb999ed3b1adc3e : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/1ed36668d412492d806f9de11eabde5c : SUCCESS in 1h 03m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0159dd7607824270a02d3b9bdee5d917 : SUCCESS in 1h 01m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8feb9579a2ae4857960df7a35cbe4c2a : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ab13652130a4ead81c56202a871e61c : SUCCESS in 1h 07m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/409157e31b9c40e6a69740e89a70938d : SUCCESS in 12m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d56753860c84495b85da67f99602b1d9 : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2fcfff5518a4452babb2e23e536db2d0 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2bc26cfbfe8845bfa2ba1b66b7768421 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/84dd71f864704011aa15fbf9516d31de : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2e3603a5be764fa7ae5b35a1a90b3fde : SUCCESS in 6m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e88cbf45ed62467bb2c924586598ccad : SUCCESS in 9m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc127e922e69401bbb07cee8ec95efc6 : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5c132e358bb495c809f5fe26dfba8ae : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eca5323444ab490b8ecc51d1648778c0 : SUCCESS in 1h 55m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ebd84c93cee49babb2c0ff22bd48360 : FAILURE in 1h 42m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e068594f620a40c39e00307bea7fd185 : SUCCESS in 2h 09m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/57e7248f495e42599e115c5ea979f627 : SUCCESS in 1h 47m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/84299513d30b410582f5dd2aac6724e1 : SUCCESS in 28m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc524e8582634e2c847cc40af976b139 : SUCCESS in 1h 15m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c5f25713f62f4f61869cdd0f87c3b80f : SUCCESS in 1h 30m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69eede45f55a46c590782590179833d7 : SUCCESS in 43m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53ccd483889c4a429a12fca4c1f18ee3 : SUCCESS in 1h 53m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcdb4ca290d24c69af470ca7014a462f : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":5},{"id":"46f08ac3800cbbaaa3b07fee91482b00290bd75a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-29 16:37:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c8e7c6c3164548c4a42a89d0d572acbd : SUCCESS in 40m 35s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/989051","accounts_in_message":[],"_revision_number":5},{"id":"f8bb510d81458db7bd28e6fe087019771a00c2a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-29 16:42:55.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_05_2026_16_35_37\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"900a920349185e3f66745aafd4e1749cb2f58ae7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 18:06:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4107/.","accounts_in_message":[],"_revision_number":5},{"id":"a1ead93eb261bc66e1a1a2be4705f314f014f829","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-29 18:14:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4283/.","accounts_in_message":[],"_revision_number":5},{"id":"93a99507db0bee96d88940d01e0d6105d4740608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 20:04:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7495d7c00e4c9092f1140747c6e5b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ad0741f7b5943ab94c56cb405fea610 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d80feed4ed1f466ca8ae026984ae044d : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3abf985328664de39d16e93695e70f44 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"55ed40f56c2b3108f3815da34685e53f0fe7f542","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-29 21:47:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-989051-5-fc : SUCCESS in 57m 6s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-51-989051-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"2b7baf4b3637cc5bc80bef091a2f37140b89b3f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 19:49:37.000000000","message":"Patch Set 4:\n\nBuild Aborted \n\nhttp://10.136.74.243:8080/job/vmstore-cinder-ci-gerrit/2003/ : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"3accc8b4b5a900af58eaaed3ea2d2bd17e7fc2b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:53.000000000","message":"Patch Set 2:\n\nBuild Failed","accounts_in_message":[],"_revision_number":2},{"id":"fac72892dd688071fcf0ad2d58cf8b6d0b58a818","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:02:59.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"f1c5f7cff0a37ed4bbd723e844e103dc81d490a8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:05:03.000000000","message":"Patch Set 3:\n\nBuild Failed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":5,"current_revision":"f340e435ea75b9a09bd29970f82606c9cf544d73","revisions":{"854befefa7c71d713520cc930206cd56ea9b0262":{"kind":"REWORK","_number":1,"created":"2026-05-19 03:34:02.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/51/989051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/989051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/989051/1"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"subject":"Always delete silent objects left behind when RBD parents objects are deleted.","message":"Always delete silent objects left behind when RBD parents objects are deleted.\n\nCloses-Bug: #2153099\nChange-Id: Ifd7d943f3d28fd072829bec351fd3dba2c16d32f\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/854befefa7c71d713520cc930206cd56ea9b0262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/854befefa7c71d713520cc930206cd56ea9b0262"}]},"branch":"refs/heads/master"},"8fbf216467c8bfd8840fc089795789b7f4faba9b":{"kind":"REWORK","_number":2,"created":"2026-05-19 19:52:03.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/51/989051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/989051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/989051/2"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 19:51:52.000000000","tz":-300},"subject":"Always delete silent objects left behind when RBD parents objects are deleted.","message":"Always delete silent objects left behind when RBD parents objects are deleted.\n\nCloses-Bug: #2153099\nChange-Id: Ifd7d943f3d28fd072829bec351fd3dba2c16d32f\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fbf216467c8bfd8840fc089795789b7f4faba9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8fbf216467c8bfd8840fc089795789b7f4faba9b"}]},"branch":"refs/heads/master"},"c06ff0b171ba25f5fbbdcca43353759a5cc5a945":{"kind":"REWORK","_number":3,"created":"2026-05-22 01:18:03.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/51/989051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/989051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/989051/3"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-22 01:17:44.000000000","tz":-300},"subject":"Always delete silent objects left behind when RBD parents objects are deleted.","message":"Always delete silent objects left behind when RBD parents objects are deleted.\n\nCloses-Bug: #2153099\nChange-Id: Ifd7d943f3d28fd072829bec351fd3dba2c16d32f\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c06ff0b171ba25f5fbbdcca43353759a5cc5a945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c06ff0b171ba25f5fbbdcca43353759a5cc5a945"}]},"branch":"refs/heads/master"},"344211d46f268c51086bb2739f5115c152df6d42":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-29 03:41:34.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/51/989051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/989051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/989051/4"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-29 03:11:29.000000000","tz":-300},"subject":"Always delete silent objects left behind when RBD parents objects are deleted.","message":"Always delete silent objects left behind when RBD parents objects are deleted.\n\nIf the volume does not have a parent, the \"fast path\" deletes the volume\nimmediately.  If the volume has a parent, deletion is delayed to check whether\nthe volume\u0027s parent has been previously deleted (now a \"silent object\").  If\nevery ancestor of the volume has been deleted (all \"silent objects\"), they are\nall permanently deleted.\n\nCloses-Bug: #2153099\nChange-Id: Ifd7d943f3d28fd072829bec351fd3dba2c16d32f\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/344211d46f268c51086bb2739f5115c152df6d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/344211d46f268c51086bb2739f5115c152df6d42"}]},"branch":"refs/heads/master"},"f340e435ea75b9a09bd29970f82606c9cf544d73":{"kind":"NO_CHANGE","_number":5,"created":"2026-05-29 14:16:41.000000000","uploader":{"_account_id":37798,"name":"Sam Clippinger","display_name":"samatwork","email":"sam.clippinger@garmin.com","username":"samatwork","status":"Garmin International"},"ref":"refs/changes/51/989051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/989051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/989051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/989051/5"}}},"commit":{"parents":[{"commit":"7f6a196c65cbf3cc4952a9e80c1a1965c6699aad","subject":"Merge \"tests: Filter out eventlet deprecation warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f6a196c65cbf3cc4952a9e80c1a1965c6699aad"}]}],"author":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-19 03:31:25.000000000","tz":-300},"committer":{"name":"Sam Clippinger","email":"sam.clippinger@garmin.com","date":"2026-05-29 14:16:14.000000000","tz":-300},"subject":"Always delete silent objects left behind when RBD parents objects are deleted.","message":"Always delete silent objects left behind when RBD parents objects are deleted.\n\nIf the volume does not have a parent, the \"fast path\" deletes the volume\nimmediately.  If the volume has a parent, deletion is delayed to check whether\nthe volume\u0027s parent has been previously deleted (now a \"silent object\").  If\nevery ancestor of the volume has been deleted (all \"silent objects\"), they are\nall permanently deleted.\n\nCloses-Bug: #2153099\nChange-Id: Ifd7d943f3d28fd072829bec351fd3dba2c16d32f\nSigned-off-by: sam.clippinger@garmin.com\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f340e435ea75b9a09bd29970f82606c9cf544d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f340e435ea75b9a09bd29970f82606c9cf544d73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
