)]}'
{"id":"openstack%2Fcinder~990789","triplet_id":"openstack%2Fcinder~master~If56ed47e9159e923943ea9a7b002724a77f84a47","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-18 07:31:29.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-18 07:28:47.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-18 08:06:56.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If56ed47e9159e923943ea9a7b002724a77f84a47","subject":"Use urllib3 directly in s3 backup driver","status":"NEW","created":"2026-05-30 05:13:55.000000000","updated":"2026-08-20 05:32:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ac84ba7bb0db9040cdd85a0756e1fd12ab89f2e2","_number":990789,"virtual_id_number":990789,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-18 09:49:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-30 05:57:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-30 06:09:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-30 06:31:29.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-05-30 06:33:40.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-30 06:48:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-30 07:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-30 07:28:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-30 07:29:33.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-30 09:12:03.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-14 14:56:18.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-18 07:31:29.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-08-18 08:06:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-18 09:47:58.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-18 09:47:58.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-18 09:49:07.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"}],"messages":[{"id":"e578725808358e2ced47cb16a83b48075fb347ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-30 05:13:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"768af087f00b4354580a993545b1ceed8b30656b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-30 05:16:49.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a04c36c9588f681650eaf5d729c5634707cebb98","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-30 05:24:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1be56fa37eecf5d6158dbd1215e7ff42b17d5efc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-30 05:57:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/0e963fb5299b4f28857576fda6ae04dd : RETRY_LIMIT in 18m 35s\n- cinder-storpool-tempest-iscsi-multipath build/1751c39d6f8b475db3a740d407f7e70a : RETRY_LIMIT in 19m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c930b3e323c7a4f40716cf3e08d5d2d072cd589c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-30 06:09:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If56ed47e9159e923943ea9a7b002724a77f84a47 .","accounts_in_message":[],"_revision_number":1},{"id":"efc59924a712eef5414b21fcbc6fcfcb913e26cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-30 06:14:38.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/If56ed47e9159e923943ea9a7b002724a77f84a47 .","accounts_in_message":[],"_revision_number":2},{"id":"ef04db37d5f34b5265f9f86053da328554ccb083","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-30 06:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/a27f621eefdd422badb4ba1fb1346f43 : SUCCESS in 47m 08s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/990789","accounts_in_message":[],"_revision_number":2},{"id":"acb807f4e4083225fd3b558288e1b43aa776d0df","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-30 06:33:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/e038c0fd8b074fb59b94f00c4101b5c2 : SUCCESS in 36m 35s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/990789","accounts_in_message":[],"_revision_number":2},{"id":"042eaef3e65a5cd8c392b181ff9d5e735ccf9469","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-30 06:48:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/ca5632276b9b431b8e73cf0b1a5b8cec : SUCCESS in 40m 15s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/990789\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/990789","accounts_in_message":[],"_revision_number":2},{"id":"0ca96b6f348e13bd4dcb0993a6cf3599e4c0c3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:28:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84280bc605b4c75b35d316bcc286b48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f31158fc59495e8eda88292b386e09 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b1fa58c44b249f48be18f5b1bfaa3f1 : SUCCESS in 8m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6696a7ab3a5e40959d01606026b26a3b : SUCCESS in 10m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/58feff7ff8d448f9be48904f53d06f51 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702d6bdcb43d41d29008a4ba0735fa9c : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/05170ecaa70a4255854cc1f4c75aa352 : SUCCESS in 1h 02m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4915acedd3164941ae7208d4cfc70219 : SUCCESS in 1h 06m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b6ab35fb58541d4ae5a37b956efc0df : SUCCESS in 1h 29m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a192684913cb4c1ab6dc1f61e716a0df : SUCCESS in 37m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995ef220c64a41609f9b514950a11ed8 : SUCCESS in 4m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0542524c85048e3aaeec343e877eabf : FAILURE in 25m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d29091e4a9aa4fd798ab19f289fbaec8 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/daecc5b03b8249cfba78646c72395889 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf32317e3faa429d85a45815843ee694 : SUCCESS in 4m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/be10a8c0567b40b593d89166835d7b16 : SUCCESS in 5m 07s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cc5c1502b8054a3d886bf0f6e167e352 : SUCCESS in 8m 45s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e75f607167144cf98041edd35263c18 : SUCCESS in 42m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/342446ce33ce47c28fa68cd63dfbb9f0 : SUCCESS in 3m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30bb9fb332294d2caf818eeea04fe7d8 : SUCCESS in 2h 06m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f36d2eef03a64a02b2d4e282de8a5705 : FAILURE in 1h 53m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01de7296c149462bb55133ddf56bed61 : SUCCESS in 2h 00m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/45d1e02c650144b89943620fd241b8de : SUCCESS in 1h 43m 49s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/faa51fd6efe347c792c2a4446fa96577 : SUCCESS in 41m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b98a31ab2864b9fb3631fb06ff3da37 : SUCCESS in 42m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9fbe9536483d437b9b7159f068a3b92a : SUCCESS in 1h 39m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0a7b39e737b84a05bff23812ec108ea5 : SUCCESS in 45m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a15d41dc62224b76b74c07242ee8a6c3 : SUCCESS in 48m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32a87120438c4579a64da4dbc03c2fc6 : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7a514e92eb89397d281b7d6f937d65c7a9914f85","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 07:29:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4293/.","accounts_in_message":[],"_revision_number":1},{"id":"cb35c744ac101e48dde40f1b37e21d3bb4a2f08e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 07:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829e92d924de46edb657bf849deb7cde\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc5252775bdd4eaa823e44a1e4ee6b14 : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/802fe8cea23c44048d0c9e253589af91 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2706868f8a434a368ae452f4cea1e039 : TIMED_OUT in 1h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d337ff23df755e5660d8098d963355bb78e67fc0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-30 07:38:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_05_2026_07_32_05\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"8f5c386e98bcc5303659d3d1db728a68f330f3dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 07:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4117/.","accounts_in_message":[],"_revision_number":1},{"id":"1a4eb88af3dbc29ac8e05273f54b0723becbe27e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 08:45:34.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4294/.","accounts_in_message":[],"_revision_number":2},{"id":"bd3145613f1a90581f98dde77fe568cf56ed854c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-30 08:48:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4118/.","accounts_in_message":[],"_revision_number":2},{"id":"889d934146c67f24b05853ce7d7260a6b920b8d8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-05-30 09:12:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-990789-2-fc : SUCCESS in 50m 4s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-990789-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"9d2540b254bab8851095a07b3bb86ce4448fc5a9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-14 14:56:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"583f026b1864a3d310298e7264e6bb8274990bed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 07:28:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0796efdde15fa36fc6e9f0dc500a06b23ee22629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 07:30:50.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"45f0bcb62273761b6a0ee0668759020d0ed5e2bc","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-18 07:31:29.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0d3d4bef4e177b4ed800c42012476c8754bf05f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 07:50:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0aed546374c4e189fce45679c880e22\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a75a097da1f442f8bc5b22452303fd53 : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f9f150683674ea7a407d6892c2d4da5 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/951e87594c534218bb583b1a9ac61f6c : FAILURE in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e58e66894ee8b3164fcf1dce446da7f4806bd39","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 08:06:56.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/bf246829700048bfbf4b11d0707d958b/","accounts_in_message":[],"_revision_number":4},{"id":"fb79b22d20ac30ca4e3822b572f81e3cfa549979","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 08:26:16.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If56ed47e9159e923943ea9a7b002724a77f84a47 .","accounts_in_message":[],"_revision_number":4},{"id":"8585959f2c1ce8903b32d564714faf63adb91217","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 08:27:11.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/If56ed47e9159e923943ea9a7b002724a77f84a47 .","accounts_in_message":[],"_revision_number":3},{"id":"1075d3e1b2df27cd175f95e37e8f8aa5c01ff913","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-18 09:14:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/bf246829700048bfbf4b11d0707d958b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bc749fd69dae42d49656ececfccd7637 : SUCCESS in 42m 56s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c89aa59e801a4e3f95872258d838e6ee : SUCCESS in 32m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d674096da4b74854aeba830bf75b4566 : SUCCESS in 42m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61fa9a0c21254e87a46e855a5fe8a0a2 : SUCCESS in 34m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/819a506c11b3451bbca22dd4a57e26a2 : FAILURE in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53aceaa13ac13341145e41e56a120961f74ae613","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-18 09:47:58.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1232 : SUCCESS in 1h 2m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/990789","accounts_in_message":[],"_revision_number":4},{"id":"c231edd3365247dd2df95eb8817def35c361b19e","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-18 09:47:58.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/869 : SUCCESS in 1h 0m 56s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/990789","accounts_in_message":[],"_revision_number":4},{"id":"16aa6230242eb532eeb10007c49b86983efac180","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-18 09:49:07.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/963 : SUCCESS in 0h 59m 39s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/990789","accounts_in_message":[],"_revision_number":4},{"id":"70faad6b882d0181f7eec720156432e9e7e12877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 09:49:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47014f9987a14f6eabc057e88b21e5dd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c18d201ab1d4838b027d21fbb950211 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0d77c74fd75433aa8bf17f843e0bdc2 : SUCCESS in 11m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe64cc851ba748f3977f3333a031bd76 : SUCCESS in 11m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cb91ce825164a8998c3155ad5041200 : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3d592ba04344ac8bc75855f7f76350 : SUCCESS in 18m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ec7b94fb831f4842b8105517dece73e5 : SUCCESS in 1h 00m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c412508e7294ffb9f159863371a1311 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a685e2f61a9e4edea55a9ac21f29229b : SUCCESS in 1h 25m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b27f4f5484c343818f8c5fc19e5406bc : SUCCESS in 30m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/288e1c5701ab4d88bc2513fa7891c0df : SUCCESS in 10m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4702de4ec1e4425abb2b47800f4341c6 : SUCCESS in 23m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5086c26d0c045f9bb02de31eeb3ce31 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b6b666b14a649929b2200ae02a0e4fe : SUCCESS in 7m 30s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/23ebd0f512014ffe894da4eea6d1d088 : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/913b54e93d9f4346875f0e0505a6c52a : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/38be2d8af41f4f4daea53b27b31b5983 : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ca71e180bbc54aae8664ec7e9f225cde : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8be38df74faa4806a6e4ceb7f96d9412 : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/434dc439c2554c81b3891a692ce61991 : SUCCESS in 3m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20af41be98f94565bc00e022bff8e474 : SUCCESS in 1h 41m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/02457e9059f74d5b9b988dd005e2a08b : FAILURE in 2h 13m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a23a4bf63cd447498054a6a5a61e83b : SUCCESS in 2h 04m 22s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c25fbaf0943f4246bda43186e5c65878 : FAILURE in 2h 03m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/70532bd46ce44270bc93e31967d34d6b : SUCCESS in 1h 51m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/aeb86a50602b4fc6b7e13db51407a48d : SUCCESS in 21m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dff315763d24414c98c349c96d146684 : SUCCESS in 1h 05m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/38d8fadee0894cfba0310c85c6e1806f : SUCCESS in 1h 06m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/73c3e818089a46d795c932d9d13c673f : SUCCESS in 35m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/791822a7bc954a7ab7ea2353578613eb : SUCCESS in 1h 56m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adac82ad296246439502944519634e76 : SUCCESS in 58m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"38966cd7054ed3c6f3740e2619ddbe53ad543687","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 10:01:36.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_08_2026_09_54_35\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"bcf3c718020012ab93abe9561384be0a2ff48589","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-18 10:37:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/c170fe9cde324db0abd6c77fa8b08dc6 : SUCCESS in 45m 12s\n- cinder-storpool-tempest-iscsi-multipath build/eb0b232839944d5fb44ab65ae8f29b3a : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":4},{"id":"6dde848cc94bbc5d4fa178a3504c4b20695a2c27","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-19 12:52:50.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-990789-4-fc : FAILURE in 13m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-89-990789-4-iscsi : FAILURE in 1h 8m 38s","accounts_in_message":[],"_revision_number":4},{"id":"4f2a6fadd7306b001463c2c571e29c7f40a19424","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-19 16:23:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5175/.","accounts_in_message":[],"_revision_number":3},{"id":"4e442ff55e381b5599fb286e552e78552b93bcb8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-19 17:18:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5176/.","accounts_in_message":[],"_revision_number":4},{"id":"1b16b20c04cb34f61e70b6ffa6f503c94ce1000d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-20 04:30:41.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4999/.","accounts_in_message":[],"_revision_number":3},{"id":"ac84ba7bb0db9040cdd85a0756e1fd12ab89f2e2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-20 05:32:06.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5000/.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d33d559c2946ca534ec63efdfc5bc1a4e15d97d7","revisions":{"df2f986a30f450d6ce5a71514d9c05547c4328f8":{"kind":"REWORK","_number":1,"created":"2026-05-30 05:13:55.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/990789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/990789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/990789/1"}}},"commit":{"parents":[{"commit":"4549a1fa7bcfb3eac6ddee3f9be8c85aa1cf8a58","subject":"Use urllib3 directly in remaining driver code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4549a1fa7bcfb3eac6ddee3f9be8c85aa1cf8a58"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 16:44:57.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 16:47:22.000000000","tz":330},"subject":"Use urllib3 directly in s3 backup driver","message":"Use urllib3 directly in s3 backup driver\n\nReplace botocore.vendored.requests.packages.urllib3 with urllib3 import\nto align with the rest of urllib3 imports.\n\nChange-Id: If56ed47e9159e923943ea9a7b002724a77f84a47\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df2f986a30f450d6ce5a71514d9c05547c4328f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df2f986a30f450d6ce5a71514d9c05547c4328f8"}]},"branch":"refs/heads/master"},"2868f12f833cddd064667c9a9afefb7c2f721677":{"kind":"NO_CHANGE","_number":2,"created":"2026-05-30 05:16:49.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/990789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/990789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/990789/2"}}},"commit":{"parents":[{"commit":"738353c242da2c9586e721f3bd6e298343e66d33","subject":"Use urllib3 directly in remaining driver code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/738353c242da2c9586e721f3bd6e298343e66d33"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 16:44:57.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-30 05:16:25.000000000","tz":330},"subject":"Use urllib3 directly in s3 backup driver","message":"Use urllib3 directly in s3 backup driver\n\nReplace botocore.vendored.requests.packages.urllib3 with urllib3 import\nto align with the rest of urllib3 imports.\n\nChange-Id: If56ed47e9159e923943ea9a7b002724a77f84a47\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2868f12f833cddd064667c9a9afefb7c2f721677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2868f12f833cddd064667c9a9afefb7c2f721677"}]},"branch":"refs/heads/master"},"431b2f3beddd9edc6f0167a5cf02b768980ddf59":{"kind":"REWORK","_number":3,"created":"2026-08-18 07:28:47.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/990789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/990789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/990789/3"}}},"commit":{"parents":[{"commit":"738353c242da2c9586e721f3bd6e298343e66d33","subject":"Use urllib3 directly in remaining driver code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/738353c242da2c9586e721f3bd6e298343e66d33"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 16:44:57.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-18 07:28:28.000000000","tz":330},"subject":"Use urllib3 directly in s3 backup driver","message":"Use urllib3 directly in s3 backup driver\n\nReplace botocore.vendored.requests.packages.urllib3 with urllib3 import\nto align with the rest of urllib3 imports.\n\nChange-Id: If56ed47e9159e923943ea9a7b002724a77f84a47\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/431b2f3beddd9edc6f0167a5cf02b768980ddf59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/431b2f3beddd9edc6f0167a5cf02b768980ddf59"}]},"branch":"refs/heads/master"},"d33d559c2946ca534ec63efdfc5bc1a4e15d97d7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-18 07:30:50.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/89/990789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/89/990789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/89/990789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/89/990789/4"}}},"commit":{"parents":[{"commit":"738353c242da2c9586e721f3bd6e298343e66d33","subject":"Use urllib3 directly in remaining driver code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/738353c242da2c9586e721f3bd6e298343e66d33"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-05-29 16:44:57.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-18 07:29:58.000000000","tz":330},"subject":"Use urllib3 directly in s3 backup driver","message":"Use urllib3 directly in s3 backup driver\n\nReplace botocore.vendored.requests.packages.urllib3 with urllib3 import\nto align with the rest of urllib3 imports.\n\nRelated discussion:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HEEWHRVRYFWQNYEYBZ5SCRAF5ODSL3S6/\n\nChange-Id: If56ed47e9159e923943ea9a7b002724a77f84a47\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d33d559c2946ca534ec63efdfc5bc1a4e15d97d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d33d559c2946ca534ec63efdfc5bc1a4e15d97d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
