)]}'
{"id":"openstack%2Fcinder~991057","triplet_id":"openstack%2Fcinder~master~Ie0a1dc78ad00d315e75e8ae2880943f39320c834","project":"openstack/cinder","branch":"master","attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-09-02 10:44:15.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-06-11 10:42:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38623\u003e replied on the change","reason_account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-09-02 10:44:15.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-06-11 10:42:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38623\u003e replied on the change","reason_account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-09-02 14:28:40.000000000","reason":"Someone else replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-06-11 10:42:05.000000000","reason":"\u003cGERRIT_ACCOUNT_38623\u003e replied on the change","reason_account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}}},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-09-09 05:58:26.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-09-02 11:19:43.000000000","reason":"\u003cGERRIT_ACCOUNT_36179\u003e replied on the change","reason_account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-09 14:43:28.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-02 10:50:36.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}}},"hashtags":[],"change_id":"Ie0a1dc78ad00d315e75e8ae2880943f39320c834","subject":"[netapp-cinder] Added support for san active active capability","status":"NEW","created":"2026-06-02 10:23:34.000000000","updated":"2026-09-09 14:43:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":81,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f32f68ef30581a5c297407e13165d679c83f502c","_number":991057,"virtual_id_number":991057,"owner":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 15:44:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"recommended":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"all":[{"value":2,"date":"2026-09-09 05:58:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-09-02 11:19:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-09-09 14:43:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":1,"date":"2026-09-02 10:50:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2026-09-09 05:58:37.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"CC"},{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"CC"},{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"state":"CC"},{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"CC"},{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-02 10:26:26.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-02 10:55:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-02 11:01:14.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-02 11:04:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-02 11:09:07.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-02 11:24:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-06-02 11:25:53.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-02 12:03:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-02 13:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-02 13:22:43.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-02 19:21:48.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-11 10:42:05.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-06-11 10:42:05.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-06-11 10:42:05.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-06-11 10:42:05.000000000","updated_by":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-21 09:26:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-08-21 09:39:02.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-21 09:39:37.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-21 09:39:56.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-08-21 09:46:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-21 15:45:26.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-08-21 18:05:27.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-08-31 13:29:03.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"a1c0844353cbf3b1f60ece0fd7c37b71e89ac98f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-02 10:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00eaab9e0084f5c6fd27a7660ab7d92750a33144","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-02 10:26:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"818ed84de7f525d9b81706a4f7fffaf4e0e5a2be","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-02 10:27:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1961784a62726210c2c452c5bd70fbf1d8dbb0f5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-02 10:55:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/1/upstream-check/cinder-tempest-plugin-ontap-fcp/001d35f/ : FAILURE in 27m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/2dbbdd3/ : SUCCESS in 27m 23s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ef0a662/ : SUCCESS in 27m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6493a28c610cefa3c59a843ca8eeb48fe59c354a","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-02 11:01:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/adcb31f4cceb4355ad95facc5117e265 : SUCCESS in 37m 14s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/991057","accounts_in_message":[],"_revision_number":1},{"id":"e599eff320d9852cb153af3ffb65e770bec7606d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-02 11:04:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a59522f6842e48b598d4d7ce4fa9a91b : SUCCESS in 40m 03s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/991057\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/991057","accounts_in_message":[],"_revision_number":1},{"id":"24ae2fb8be6f32a3a2cdc1fd307f7871cd7e47d8","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-02 11:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/f73b0da551ec4d5a8e55e30a9881a2e4 : SUCCESS in 45m 01s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/991057","accounts_in_message":[],"_revision_number":1},{"id":"61d5af0e04101b8843b9bf841045de826ce826fc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-02 11:24:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4141/.","accounts_in_message":[],"_revision_number":1},{"id":"4c860e879c6bba9730fb95eafe1da24272d5235b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-02 11:25:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/d772f50688b3492eb11d1836fc841c40 : SUCCESS in 1h 00m 50s\n- cinder-storpool-tempest-iscsi-multipath build/58781a5105844f449889a93fa4b568b3 : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fd5a518f71286d927f61472262beec2bc59ba413","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-02 11:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4317/.","accounts_in_message":[],"_revision_number":1},{"id":"1ebcee3110f6600c95489daa9806379621f439d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 12:03:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adcf453b6814dc3a5482563179da360\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acbda3f21afb4e9fb9e6d65c08d0363d : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/80420cba07f34614bc833a2584a8aaba : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b0e6452576a49a99db42f9b07b6a6b5 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09e2b65585eb53688ca930057ed58d0965dab07b","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-02 13:07:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ee8d3833c7ad948077c5cdb2779474747912022d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 13:22:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58f5f15bf77b4adba517e5dc9407c9bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f51759748141668baf5562ceb01add : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98cdc82959974818a1faefc10775e81a : SUCCESS in 9m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7806c738308415c80329a4df29fb044 : SUCCESS in 9m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9da7e6ec09434656aaef4cd8c6390c67 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc89d86edb3a4a0e9db2ae7ffea0a6b9 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/1154c5147eda41c3a50bda21b9f6dc16 : SUCCESS in 33m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fd0396b34f444608f10bde1890207fe : SUCCESS in 33m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16c3ea3397354790b389c7240043c049 : SUCCESS in 1h 41m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d740e146cfe54120937d91220ffcd4d1 : SUCCESS in 34m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7edd794862b64208b28541f76e92746c : SUCCESS in 8m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e413f1f1eeac4486bc9eab272c17c4d1 : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d92f7f0b9e2c45e98e35e92f8dc9bebb : SUCCESS in 6m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4806f23a428f43fb9a229f28c79159ba : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3d31f1ba026f47d19c2cf490b59d76e2 : SUCCESS in 8m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1c4b59178271473d809bbbdf3e618ba8 : SUCCESS in 5m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5a2acbbf0aba41aa945996913cc4e591 : SUCCESS in 10m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60c2bfdd8b004b708bdc49b63df92506 : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba83a37d872c45dea69721aa7b9b2044 : SUCCESS in 3m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7fd8aaf1f9e2439db3f6843e29418082 : SUCCESS in 1h 39m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d6406ea9c364905a271ccde6dbeffff : FAILURE in 2h 53m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66d819e70ed84e10ac5df5f1b5c361bf : SUCCESS in 1h 55m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/87d5ad3487414dd8b63aa0b23e234fa2 : SUCCESS in 1h 33m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2706c60426774d47bcecba770e87d3b3 : SUCCESS in 36m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/57577c7ad313455b825db7533aa17365 : SUCCESS in 1h 27m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/387e1ae6a0d24c2eb2cbfe622ba1df4a : SUCCESS in 1h 36m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/00aeda257b364a8ead0a5140d86c9068 : SUCCESS in 24m 35s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20997c5366824cebad4fac9536a78052 : SUCCESS in 1h 44m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50eb64fb81f143aba3407d760131bdc4 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9ed668ef25888534843390ac50b806b6dc0d8cd3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-02 13:33:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_06_2026_13_25_53\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"588823eb27de30dfc5fd4f5ced90a87ec63b4ccd","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-02 19:21:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-1-fc : FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9af0f258d089fa124cb23169f4566f237cbd31da","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-04 04:18:59.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dfb0297a31a451a485ec2c480d0ea00bf8073364","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-04 04:19:28.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fff80962faa73209b1e521ace902cebf2e3a62f6","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-09 05:45:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"26b5235fc636a73fd4e10ba07bff6157f2b54b45","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-06-11 10:42:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"591368495cc5f7b7cd5c8a36d750b1f01c0a6db6","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-10 15:43:56.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ab1854df9439b0a4b2633abeca19901c439c8f80","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-18 10:13:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ffd38dd2bf6461d55a1b4b0241b818bc6c3bfe86","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-19 04:37:55.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"94c18d7942cb270a518638068fea4a5c0999bf3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-21 08:51:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"976c6c2ec4dcb9fae074a401f5fcdb26788b7c62","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-21 08:53:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ebe247ae9e02ec472f8751fbdd21aaff44d3bcba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 09:26:09.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/548a2601f8f849de88ad220d32b7618a/","accounts_in_message":[],"_revision_number":2},{"id":"87be24b5a21e2d0d135e249cf88c1013efbc5ac9","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-21 09:39:02.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/905 : SUCCESS in 0h 46m 43s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/991057","accounts_in_message":[],"_revision_number":2},{"id":"c3d246d503c6a06ece5a40405368e7097b3fc158","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-21 09:39:37.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/998 : SUCCESS in 0h 47m 25s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/991057","accounts_in_message":[],"_revision_number":2},{"id":"0d7c032d47fcb8b96508d8a7c8f75eb807cd4782","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-21 09:39:56.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1267 : SUCCESS in 0h 47m 40s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/991057","accounts_in_message":[],"_revision_number":2},{"id":"dbc3f02f37f3c195d831648698f52d61379783bb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 09:46:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie0a1dc78ad00d315e75e8ae2880943f39320c834 .","accounts_in_message":[],"_revision_number":2},{"id":"a55cf0b3d048762fb6eb6d4b49e9d6b6106f47c6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 09:54:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/548a2601f8f849de88ad220d32b7618a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c61a67f810fb45fa8ac3fae11537cb72 : SUCCESS in 40m 52s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d6205ae5ad734444bb5b9d48293f9a7d : SUCCESS in 31m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/35aaefc165374287b22243c8080c0ebb : SUCCESS in 42m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd51af8a431c4407a0edf3689be5742b : SUCCESS in 32m 04s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8c2e0bb03a814719b245aff23de3a65d : FAILURE in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"edd50d8b0b1f38d87922684702702f6fa7ed5999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 10:37:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3553a6c16be543809e9d95826b71b43b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/70d9a2512ad348a785d995d99db144eb : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/feeb50bbb6f94758839af55c031bd80b : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/09268be037954d27b93faab80c999c01 : SUCCESS in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c98965e6d5a060afb37eec5d3060735c62c32f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 11:07:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acf42ab40b7f4180a22da9ad2f6d1e46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dcb79b651d041e48cfc23902c55008f : SUCCESS in 7m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edfdf8b2e05d48f7816fa0ee570de255 : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b776280c205a44daa4043172db48b55f : SUCCESS in 10m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24c3f367bbe34842a5bb0149fad3165f : SUCCESS in 15m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd0236e397e46ffa123b63c3a3fa89d : SUCCESS in 14m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/3b7e08fef20f4c97978e6ec6a201f579 : SUCCESS in 1h 08m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43d61403351d4683a37f2a1034b76304 : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/820de5e510df4c1e99e1fa555a79e916 : SUCCESS in 1h 18m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc88d459b9bf4a52958c31b260c46b89 : SUCCESS in 1h 04m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09f5fd8b9e8147618b67e707700d189a : SUCCESS in 9m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a32e836d523f4c1bbdbfc01cf6d156f0 : SUCCESS in 22m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13425b27d0b2492880a42e08279ca178 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c02ae240c51a46a5b255d9767b613eac : SUCCESS in 6m 41s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/7de27a8a961f4453b067d4484aaf94eb : FAILURE in 12m 43s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/85ed9728d70f4448b413912c8962774e : SUCCESS in 10m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0ce13ba60bca4249ad042a921a5740fd : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e3331e784d3a45c18ffaa4a821fc011d : SUCCESS in 9m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13aecd3f0caf47be9d2e5d7dee38b780 : SUCCESS in 35m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6f8f3ae554be46ea95552c9574507cf6 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c88c39c1f1b4f6ebd76359cb50b6756 : SUCCESS in 1h 37m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28cbb8b87016416f9404309d6914fe3a : FAILURE in 2h 03m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/245e2e4da6a4425289ccd8e08781f564 : SUCCESS in 1h 38m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/99137630c4f24a11a7b7ff1d6cc9ab6a : FAILURE in 2h 11m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a41104a9b0d4426889b12324f1527e71 : SUCCESS in 1h 33m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/03666a2820114ab38d81b87eb5ea29c7 : SUCCESS in 33m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1be1e93bc214716a3640ad2b304a317 : SUCCESS in 1h 08m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d882ed3ca064aa3935c6cb8e633b6b7 : SUCCESS in 1h 31m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0ccfe1d228d14a53af0e7436e833cced : SUCCESS in 41m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c01e71c432484002875ee2dccf337ee2 : SUCCESS in 1h 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46990f3dfc2848498c6ed83ea7731f37 : SUCCESS in 55m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"6cca6c0acc7db66f090e59b5c8507da1e07245bb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-21 11:10:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/f1886f5df5394187b9353e16ac0165c7 : RETRY_LIMIT in 22m 05s\n- cinder-storpool-tempest-iscsi-multipath build/50032c3beaa64a158d0b7102ab446840 : SUCCESS in 2h 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"154369a115b8f90eccdfdd2414e735312d50ace0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 11:18:13.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2026_11_11_16\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"246c449dadf0a6d73c5fc3e1986e5da70e64c0e5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-21 13:35:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-2-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"261f685f74ee75fd8531371dbcb3e3a8c1a37cdb","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-21 15:42:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cd077b88aed5be007978dcddce2fd0bb917a5ff","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-21 15:45:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d810a0a9efd45cb71e58890a72fd81ff4bb58a1e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-21 18:05:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c85e12772cbd3912c32e6e2d9cac59dbfe54dee5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-22 21:57:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5212/.","accounts_in_message":[],"_revision_number":2},{"id":"f72caca26ca6d85f331d6a9865c645bb69397abe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-22 22:42:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5036/.","accounts_in_message":[],"_revision_number":2},{"id":"900d30a6307b340e7ddb245c0f09ae749e86b1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-24 11:18:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8fb78f8f56c9083bf7c20973e7427dbb1083bde3","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-24 11:20:08.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"266f2bee52e216e54fb126431feb9ef8c186c5b8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 11:28:02.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5227/.","accounts_in_message":[],"_revision_number":3},{"id":"d4ed1afb19a7cd6535f9a4a2446c9d774ba00697","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 11:33:05.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/99190d94e54f4dd2aea169656e39140d/","accounts_in_message":[],"_revision_number":3},{"id":"28aa4527b81e2286f00023ab8229bc003cad96aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:51:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a16366810464aecaa7af463d7cda6c6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac2383279df44629ade01614da086abc : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1989a190e9bd42619ae9db706cbd34bc : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6886087d69fd4e5cbbc0c12145ef1703 : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b26f9dc5b5a6cab4f0b3fb50df0f9c996130c840","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 12:01:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/99190d94e54f4dd2aea169656e39140d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/07dc5f6a205c4fafb28107d018531aed : SUCCESS in 38m 00s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/61694b78da9f4ef995afda2c2d73f7ce : SUCCESS in 32m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/248c70b998164862aafe170ed21e0be3 : SUCCESS in 42m 37s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ceb9c82b65394399a520db8a3d24e989 : SUCCESS in 35m 16s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/876ba051c9764cea8a77aed14d7e07ad : FAILURE in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fe8c6a85847b14ba3fd6ae33fb21b5a9ea5d098d","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-24 12:06:12.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1008 : SUCCESS in 0h 47m 4s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/991057","accounts_in_message":[],"_revision_number":3},{"id":"27cf47c07d34e95a5c1b8ade4739994bd8da3361","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-24 12:06:59.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1277 : SUCCESS in 0h 47m 54s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/991057","accounts_in_message":[],"_revision_number":3},{"id":"e75e8ba4b914d83692cc35e1497679890af6cf62","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-24 12:07:56.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/915 : SUCCESS in 0h 48m 54s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/991057","accounts_in_message":[],"_revision_number":3},{"id":"6a2c199b95b8372fc58671dc5cd70e426f3feaee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-24 12:12:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e9b95978cb1a4b80a15fe6e89efbb103 : SUCCESS in 50m 03s\n- cinder-storpool-tempest-iscsi-multipath build/cc2d876926c048759fbcd16c08f69d2d : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f0f2379f3849b9a1f6a781cf9206934c1d6455e7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 12:18:03.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie0a1dc78ad00d315e75e8ae2880943f39320c834 .","accounts_in_message":[],"_revision_number":3},{"id":"9497b5d98b21d534c402f4121499dd40096beed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:33:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cdc809929343c7a51680d82fe1c0d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc55e8c2d0a949dab3e4c18a64fcb4e2 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e638342953f4befb7aee19ccaaa8c9c : SUCCESS in 11m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b71ef6025b18443485ed390e3b1ae540 : SUCCESS in 10m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/583f470b6fc7449189f3dfdaa6eadc39 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ae506a809747ffbd24d30585fc8385 : SUCCESS in 7m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/81de612f895546a298d2dce4ff9f476b : SUCCESS in 37m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37019c9d2a874c2d958345c71080de19 : SUCCESS in 32m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77803ccf3d41421a9bb7af751936060f : SUCCESS in 1h 30m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01877ad8acde498690345da401a7e057 : SUCCESS in 1h 03m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/269a412edb194f8193c45222361268a9 : SUCCESS in 9m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4473918ac9fb40018ff33e5de5aee526 : SUCCESS in 24m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22f905e7c54c45f4b7e01cda6066c217 : SUCCESS in 6m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e116c715a52e4ab48ee3a4b351dc79c4 : SUCCESS in 6m 44s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/059df84a28b54ea6a353464b76f90e85 : SUCCESS in 6m 52s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/602c6e13a30943e38e0ab538ea07a0c9 : FAILURE in 8m 36s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b5dcd3de85264ec2b62db55cd81b9ef9 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/57b31feeae8e4776bf76e8f5172ef128 : SUCCESS in 6m 09s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/004acfe5555e431a8a1e5d7d0048d69b : FAILURE in 5m 31s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5061802d109443619196c4c3ab57b8ab : SUCCESS in 9m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dcc16eadc5a1405984006c141479dc96 : SUCCESS in 34m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e8279f5a3580453aa95f29b5bd23ff93 : SUCCESS in 7m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2a7db504f00424193c5b1d49e91a105 : SUCCESS in 1h 17m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c220e7e296704c978cb7fad68d3c8f3a : FAILURE in 2h 01m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d66a1270c184fb49da7f9b7e45f4a01 : SUCCESS in 1h 51m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/203f86f46987465490e03dda134dc153 : FAILURE in 2h 10m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d2891460a390429e8fb72de8287a9860 : SUCCESS in 1h 51m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db0d8452b6b74603b8bfa132a178dbe2 : SUCCESS in 37m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efc5a6b50e2a47d287efda0425ba454a : SUCCESS in 38m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/738e5a97080a4254b340ce242c5e924f : SUCCESS in 1h 43m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e6172be402fb478287c8529853ab1476 : SUCCESS in 50m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ddf3f1d1e884b0ab6b8f97a6f9135c0 : SUCCESS in 1h 42m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9186ec218244bb6a7e2888bd2cfa359 : SUCCESS in 33m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"83b700418d82ec48092b04bc3d8a08742f0fbbf3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 13:44:34.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_08_2026_13_37_13\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"44b7c4054405ceb2732f38c73593ad20c82c7c91","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-24 13:58:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-3-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"670bc8c5ea3ee853da3eec87fcde74ec3be221a7","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-24 14:18:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"92f80bb6717aaa77fe3a9f4a99d0f033f83ad0bd","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 22:55:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e463c0125b19e1791ace54c4effef11f78f95547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-25 15:08:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"eb533b87c62c530458293d584ae497e25d0be51d","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-25 15:09:08.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"08687c89f556c852913e62510329cc8d14c05e38","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-25 15:09:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"433f6a4c68824761f05812cfedd361871ee5c572","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-25 15:10:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91febae6a3032602919215599975419692f5fd38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-25 15:17:06.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5254/.","accounts_in_message":[],"_revision_number":4},{"id":"496961b2336b3d50776a6fbc9b249ef0274ca6f5","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-25 15:20:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"03c81fc8d551a83b471faef9f426e9a1618e4c2e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 15:28:12.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/804be4e639b641f5b7b11e94fe7dcccb/","accounts_in_message":[],"_revision_number":4},{"id":"73545891a232c02c5833b6b846ebc9ed05f755ba","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 16:03:58.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie0a1dc78ad00d315e75e8ae2880943f39320c834 .","accounts_in_message":[],"_revision_number":4},{"id":"946425bc547e57b724d3a6cdc717bbd35ae4e27d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 16:05:43.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/804be4e639b641f5b7b11e94fe7dcccb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/35f2efd872c645e782771ffe3e08d60b : FAILURE in 47m 47s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0271d9652d5f43e8b6d1f2cf5bbfeab8 : FAILURE in 46m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8623968308540b7a98d3e787de25f61 : SUCCESS in 49m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/865c120fff7c4bddaf4e75d99d785bc4 : FAILURE in 45m 43s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e296f103c1af4ea9ac631c9f990e9355 : FAILURE in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"48f1abbf7ecad395c958a5a563b0bc6421475265","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-25 16:20:07.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1033 : SUCCESS in 0h 46m 11s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/991057","accounts_in_message":[],"_revision_number":4},{"id":"78c3acc903ab9ee38058b7be0aa4042c70378f1a","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-25 16:21:09.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/940 : SUCCESS in 0h 46m 42s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/991057","accounts_in_message":[],"_revision_number":4},{"id":"024f79d9faf01fc9924abb1fd77e70fee34506d3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-25 16:21:57.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1302 : SUCCESS in 0h 46m 56s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/991057","accounts_in_message":[],"_revision_number":4},{"id":"cd86b49236e84d95a75c7307481e5a41725f80ad","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-25 16:45:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/eec5a59b10e0494092958a0ccdb5abf4 : SUCCESS in 46m 43s\n- cinder-storpool-tempest-iscsi-multipath build/cd340dc5d07d43dd9138a0d5d310ba76 : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":4},{"id":"1e313282f3ba8077c975d30f44288ad559095b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:12:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4655299757146fd92c442a427d973db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c3a2c215004ee6aeb62b3d7a3e3493 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5591668e116246d1a44f425559a647c0 : SUCCESS in 10m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f72a1c0a5754b6c9a8ed52ca346ebf9 : SUCCESS in 11m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b1ad1bb489b409692c8a7e1c7c07092 : TIMED_OUT in 41m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054ebef173a84bad815fc0ffcde4b48b : SUCCESS in 16m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4acbc3d97c0c4324a9a88f305276abd4 : FAILURE in 26m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49189d1c4554491288add12c9e648ac2 : SUCCESS in 1h 00m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38c737baaf8a4beeaef3316528f6495c : SUCCESS in 1h 27m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f41a015934145b2a96e0098bafe7cfc : SUCCESS in 42m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3cda12c1ba04c378f9fc0feff0dcbb9 : SUCCESS in 9m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bebf859e3d464442ae28f0c103359826 : SUCCESS in 19m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c4f33d340ad94a7bafedd1839d811a1b : SUCCESS in 5m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f8b1b6a50154fdebb0ca462b308018c : SUCCESS in 5m 59s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e29aaf5f7c8b41429f2dbf7b33ebc108 : SUCCESS in 11m 12s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9a0a8e5bd69f48f2969507d075b433fb : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5f54431e547140368ef32792f425cd4e : SUCCESS in 6m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14d9b72bf0554b2d86616b1203506ad8 : SUCCESS in 8m 11s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8fb3c6910bb64382b01d96f4d112bcbc : FAILURE in 4m 09s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/205f72f7bd804b7a99b93927fec68672 : SUCCESS in 9m 52s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39c4c8d8f4e04285bc5e0d866a81c186 : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f978bf347ad4432896a7437263f92e6 : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5cfb0853372840f2ab3c1d0f3de68d01 : SUCCESS in 1h 39m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8891aba2a71348848a024070088bc03e : FAILURE in 1h 56m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b7129a659aa444488aea2b4d3556878 : SUCCESS in 1h 49m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/7b8ed9f25df441e5a81ce15a13dee468 : FAILURE in 2h 47m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/55a3bd1a6e104c768c53ee998621d9d9 : FAILURE in 36m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e50b99d7cb6e442a8cd8c47ae352d304 : SUCCESS in 38m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df71fbef29134bd98ff7c3a3f57948bb : SUCCESS in 1h 17m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2554fb7688c04da097ce46163f34dbf2 : SUCCESS in 1h 21m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4cc0ed1551324192b411f519f7a00fb6 : SUCCESS in 41m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59fb92f1ec5e413f900710bd9564e4ca : SUCCESS in 1h 33m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82465fd93efc46238f9c3c9e50c00d33 : SUCCESS in 44m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"1c4820ed5749f23779f89a99081d8b02f51eaa95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 19:44:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f1a55abba4448f0aa2a2f38267989ab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2f97f560907b4b5183860e7463263248 : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8e6506f6fcb431d9a593e9a81f4025f : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/89e70200dedc4d93b8fccac2c213d8c6 : SUCCESS in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"940aaad2f5bc212e6caba58bd0cea585037fab14","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 01:01:10.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5051/.","accounts_in_message":[],"_revision_number":3},{"id":"a2a777f28319a093b5687296a7bda292efb94427","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-26 05:41:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1724f08059c51bd5a5db2480bc1cb6cf18d9abd4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 05:57:42.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/196f9dd902464f39bb104eb88bfa408f/","accounts_in_message":[],"_revision_number":4},{"id":"aa60237e4b91a0e42443e162e6083704f0b8b40d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:26:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e650bb543cc4b1b9ce1b378bc21d7ba\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8cd8ec74a1fe4d88934037b49a4b5aa3 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/34dcb21f4d0d4eaab7fc0109c31fe6e0 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7ee60609d92d4fe2ace1e523b2feaf60 : SUCCESS in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9827a737a2682dcab66ec63face273f00eccbd1e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 06:31:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/196f9dd902464f39bb104eb88bfa408f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dfa7bf8df340432fa8a8155450d2e46d : SUCCESS in 41m 40s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/165aa0ba37b4454f96af3e412c936ac1 : SUCCESS in 34m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a423b802aa104522bd288f32fb6e3c88 : SUCCESS in 49m 46s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98c6faefee1f49248d05ed6c8399ff31 : SUCCESS in 35m 57s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d56e1b83afa141c6abdd49728b05895e : FAILURE in 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cc7cf439b382ddf2cb167fa32b81bab70e0d2cb0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-26 06:32:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/fae1aeb013ee4356a91584b86cd881e8 : SUCCESS in 45m 20s\n- cinder-storpool-tempest-iscsi-multipath build/bfc38bf8eeb5434aaf400a1573b317eb : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":4},{"id":"715b273e5f8df5672ef34c59286ac46af2576391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:08:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/058f63f3cc264102afe2234c5d4aa81b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430969646f82431083ef4a6e88a18496 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a876ba2098674a0488218bf8c0a62ed7 : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/615b4483cb154d43a41cc2f2298867c8 : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a07686eeb7e43f68399ab08d7782cb0 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2d7203b2664f518616d4a51e0a500e : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/fb31c594ed904eb7b43e2a0f9414598a : SUCCESS in 1h 06m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e475a399addd4111b0183f909b06daf4 : SUCCESS in 1h 16m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5a228d05e13458ea16917cb7e5d446e : SUCCESS in 1h 34m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c180a4441b12459f909ebcb6a0bc244b : SUCCESS in 28m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/454fb8a43e7948e49885972eba217822 : SUCCESS in 5m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cfca948ee1ec470da0849f789b898254 : SUCCESS in 24m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6715477a45844fe486f309e6878879cd : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bb5a9f2a6677423faedbe0b706867e12 : SUCCESS in 6m 04s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0ef759e1c2ee4fd6ace40befb2811392 : FAILURE in 9m 07s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8cf30cc8d5b24c14bfdb508754738ee4 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19ed8d2d6d6a49ad8bfab14b9e672ed9 : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/89e64033bb154f08aacd097a87b4bde2 : SUCCESS in 6m 06s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/30b3de32683a4e55bdb3925d798ddfa5 : FAILURE in 7m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e3cc20ebe6f647fb95a1ef69abbee859 : SUCCESS in 8m 40s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/32c2f0071f264bb7919feeb04897b913 : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/45bd64ed4ad14577a058104ed9cc8533 : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/38be79cd11ff44cf9fa6656a499c9294 : SUCCESS in 1h 30m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/04e5ea8efc5d483f83e16e592818d7eb : FAILURE in 1h 03m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bc564ec6dc148a2b121c60a818e4d8a : SUCCESS in 2h 09m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f17e2e3e5d6f4e03884ffad1c03c7d8a : FAILURE in 2h 21m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fccdc803fee84cd8bf30ce8fb8752b8d : SUCCESS in 1h 43m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6e91db0cc1f046c6a1899fdcea585038 : SUCCESS in 33m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d00c854c313424d84a6c0bfe8a75fef : SUCCESS in 1h 14m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca835a170f214bbeb18477f6ad8b7996 : SUCCESS in 1h 40m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3bca8cd0b79d4cff8f52b049b98506a8 : SUCCESS in 27m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/278f05c4140d4f7ba057c9b951d15e01 : SUCCESS in 41m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be360b939162400f8a4469f957a96611 : SUCCESS in 53m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"8a6d9cc7a89563b08138695ef6e89cddeecbce28","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-26 08:15:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7edbc307386ac36fcf375ad305c4b05711c80298","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 08:19:20.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_08_2026_08_12_28\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"d8fe1757ebc681174d16b059bc5bab45a769e7e4","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-26 11:54:16.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dadf9615eaf6a00ddeb663044a97cce05d223e0a","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-26 12:18:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"12d5f7122aa331c867acde1f9aa4541aba89c01f","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-08-26 12:59:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f9242c8a831fd89cf7905b12cca136ada272bef","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-26 14:01:02.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1035cf2c08d2cab2baf8363baeae6419a187cbaf","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-26 16:37:42.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"227f4348cc5c66de590699cf38babf569420b5a4","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-26 23:00:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f131c2a96010d6ab0c1438ce5578b22dcbcb71dd","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-31 09:45:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"102f91c8c4a8c09a075c75d66ae95589265d827f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-31 10:17:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/4/upstream-check/cinder-tempest-plugin-ontap-fcp/0ef1e5f/ : FAILURE in 26m 53s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/e205e64/ : SUCCESS in 28m 52s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/9f0400a/ : SUCCESS in 27m 51s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e84372d0618ea03081fcfe6db0fe1bb3d660533","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-31 13:29:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f067fd484550215567f3f4591099f11504d58f7","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-31 13:49:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80858303f6ec1ddfbb82da608703d9099018f0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-09-02 10:44:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"a987c7fd52a5783952c825af4a09d2560c482235","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-09-02 10:46:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a46c201a2acba64e50ddd2a828cd77851cef876","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-09-02 10:47:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"54ebb0d782f51184c4a6922774eec8b9395d20b5","author":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"date":"2026-09-02 10:49:07.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cba2aab8ecf11b25b7bdeb5bd183e39cde6381d0","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-02 10:50:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"78f8caa322d5aef1b0e5305114337977f6c52e5d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 10:56:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5b91a3de00a74475afc53f1debf13b5f : RETRY_LIMIT in 9m 04s\n- cinder-storpool-tempest-iscsi-multipath build/87eb54bf1e3a43eaa8bd58fa1121fd13 : RETRY_LIMIT in 10m 07s","accounts_in_message":[],"_revision_number":5},{"id":"c546ccb44530c11dd98021ed27d5489fe23a77dd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 10:56:33.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ie0a1dc78ad00d315e75e8ae2880943f39320c834 .","accounts_in_message":[],"_revision_number":5},{"id":"2427e22d3022d68940705a3ef4e0d3435f5587e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 11:17:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bddfa213d9ac429eab71d4761b44ecac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/82e70f1fab604871980aa9328342f4d3 : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdcf64f0a05d4097be63b260f092b102 : SUCCESS in 18m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df8efa159cb34f76a2676e7383dee0f5 : SUCCESS in 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"63b5c49ee8ac80709aeafceda87daf8916587fed","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-09-02 11:19:43.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9cfcfba4f3bf1137627adcd10b799f3c26f5148e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 11:44:25.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2029818529f54284865d2c87f8ff8d52/","accounts_in_message":[],"_revision_number":5},{"id":"251084681dbd7fc60a588d5112fa2a642bce66dc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 13:04:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2029818529f54284865d2c87f8ff8d52\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9c9a8000c40488c8346d151a9306c2d : SUCCESS in 43m 20s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/64aaa1179f0944a1ba602e27634ee331 : SUCCESS in 38m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9093f22f8dff42e1a2ef820cfc9fe8cc : SUCCESS in 52m 41s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5e62314b94a3435d8b46a06c2901008f : SUCCESS in 34m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20d186d79cf348bea8425919ee3d8fe1 : FAILURE in 15m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2c91cc06796ee6ef808a9e03976af369325598f9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 13:23:28.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1144 : SUCCESS in 0h 46m 1s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/991057","accounts_in_message":[],"_revision_number":5},{"id":"6e8492fca84d2e3e969519582f18f0f9a13f24ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 13:24:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0b029ef690643269954b09915faff37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8af0312058406bb9638a701a8420a1 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91a8088153894d5c83b4c6dade7ed919 : SUCCESS in 11m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53d7f7575a5d451f9cfc8392dc9dbf6a : SUCCESS in 9m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d869d017465f44d3ba09cb375057c30b : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7997fd189834d8dbc97d0666e056e03 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/8823150ace5b4031a594ea1b74540747 : SUCCESS in 1h 05m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00726933edbe47e9b1cc02abba8bc565 : SUCCESS in 1h 20m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/58dc95398b5e4e0abdf67be67b76eabe : SUCCESS in 1h 09m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42721ecd2338485195bebd7ec91c4a30 : SUCCESS in 36m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba16f7a9d7ed4603a81a73f22a7e422d : SUCCESS in 8m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/509813d79ff44ceba78ee983535c4058 : FAILURE in 20m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8d4a0ba5ac7d42b79a11ea4d8e6890c1 : SUCCESS in 6m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3bc9b572dde64f83965cf0560eff5ad9 : SUCCESS in 7m 34s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/01a8439b27264580a81eed393f48c39e : SUCCESS in 10m 14s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/85ae22dcee7d42a9974e7a80d66b75a8 : FAILURE in 12m 15s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e815ad52fe2343aba047c4fb08b957d2 : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/10e7939047584ab38f05736443a4e607 : SUCCESS in 5m 55s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fd4491121588480e8518fbc2b700c3a2 : FAILURE in 4m 19s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d12bfe42c97d40fea952aa71635981b7 : SUCCESS in 9m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ab5de15d3b441b786b23eb8762726bd : SUCCESS in 35m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/487db1deb0d44fafb0dbcb79b2746571 : FAILURE in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8ff4d60669614c19a150d8acdcfa6144 : SUCCESS in 1h 36m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/23abc644d7654d8aa95ad399da19e327 : FAILURE in 2h 05m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f6a0ddc903914476a1fc39fc4c915231 : SUCCESS in 2h 00m 20s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/619f710c43a1479792bb214becd238fe : FAILURE in 2h 25m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/289c3e2282cd41ff8186623df1abde33 : SUCCESS in 1h 34m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/db65a1079b8d430282cb2c46e46d405d : SUCCESS in 35m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dbff914409d14224a419ac1b16fd3db8 : SUCCESS in 1h 18m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/21467b1d9c944ba9b07cdfa2dd62c3f6 : SUCCESS in 1h 26m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b7be0b0a6bae493ca14771cdbbc4df61 : SUCCESS in 48m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a311321a40e4001871b9cbb672aa3e1 : FAILURE in 48m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23b0363bea3349318f7a2d645a1b7d3f : SUCCESS in 1h 03m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"8974d3726c2b5e9d50a2d871ab3455c1852a92cc","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-09-02 13:25:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e80a75874cb30e35cefbf5da39320b28baf2111","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 13:30:50.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1051 : SUCCESS in 0h 45m 20s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/991057","accounts_in_message":[],"_revision_number":5},{"id":"dccad057b878835438f8bed0c8d77a4c420e785e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 13:31:57.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1412 : SUCCESS in 0h 45m 26s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/991057","accounts_in_message":[],"_revision_number":5},{"id":"341ffc2b8789d69f7fd875b475851eddec4fdae1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 14:28:40.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5373/.","accounts_in_message":[],"_revision_number":5},{"id":"0325817208f18edf1833e9c4947468b4fd3c0bb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 14:40:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3859a085d3640ccb298dcb6dc828dbf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/591ab3359e49459ca25e31fb2bbb6ea0 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f30e7fb9b6684e809a6a36197403cd78 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2731a0ed8f4c4d489b014670228b166b : SUCCESS in 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc06133812d95aac5d9d69a255db6dc66788d25f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 15:35:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/38d1f616aae3433996443963eb1219f1 : SUCCESS in 46m 18s\n- cinder-storpool-tempest-iscsi-multipath build/4a988b0715fa45e7bd63bb2ed5661ae4 : RETRY_LIMIT in 9m 50s","accounts_in_message":[],"_revision_number":5},{"id":"b1206efbcaa13857cefaaa32c5fb7d64b9d2cbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:44:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c663fe0f88f431483b12f3e619793d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/801222f10a1c48deb5a55b6d82bfab85 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e4a10ed599c40d7ae6b1ce231e9fd4a : SUCCESS in 9m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a40675632f9a4815882d30318d037f48 : SUCCESS in 9m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a47812569d5466788fa262598580ae4 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b531c15f2340a18fc059226dc5204a : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/e059182da1c04443b597b7285305681f : SUCCESS in 59m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c08f02305044dabb937d8ec829e73de : SUCCESS in 1h 11m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccfa0c144a3641ee9445629b581dd55e : SUCCESS in 1h 23m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/044613c411aa4221b98a72e85ce91eb0 : SUCCESS in 1h 05m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17dd8696c02c47cbb627f16ea5997d76 : SUCCESS in 10m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f67101fefacb497a97d9fee3271de34b : SUCCESS in 18m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/36a17627b36c43e9a959e1ba37152392 : SUCCESS in 4m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e869a2733a1c4f9692ba63d217465db6 : SUCCESS in 6m 53s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/aa1f6d8ce3b14682a604414758bc2011 : FAILURE in 12m 20s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/128ebf78b90a41368876f5b275842a2f : FAILURE in 8m 41s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/836a30fdfb354d6b85e90f6c5f138bfa : SUCCESS in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b11ccb4a586f4399b4a9156f30fe96c9 : SUCCESS in 6m 03s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9b83c0396bfc4a1ab58c82820a484d32 : FAILURE in 5m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/94ba00d0aec14fdbbf997cfe73c55e2c : SUCCESS in 10m 52s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/61808960f8b84b40bfced7e61acc1de6 : SUCCESS in 44m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4dec8be97ad44d67862c0d132ac8502a : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/457315435fe04bffa66847e6a16e93a7 : SUCCESS in 1h 13m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3aa188d110ed44a7b654a8ae5656dc57 : FAILURE in 2h 00m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61460a508e714557b55f450b527bbb6d : SUCCESS in 1h 49m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f0db1403ae1f41f0abd3b4aeb2c6f263 : FAILURE in 2h 05m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/95ab1a9bf6214b1e9380f1a272b5bb42 : SUCCESS in 1h 43m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/62955597d7cb449b89acf63005a9354e : SUCCESS in 37m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d311f6dc01d9473d904cb5c79673d818 : SUCCESS in 1h 13m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/412c59509e8044b78549426f0e6c7640 : SUCCESS in 1h 33m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0c25a2f2351048f5b4b6866d739132d8 : SUCCESS in 39m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b650ce4898224f29ad06c9f1fbb370e8 : SUCCESS in 40m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c32d3cc68e044199b0fc3832a041da5b : SUCCESS in 41m 38s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"482116115bbb0b395c3b7d4facee40bf295faa4a","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-09-02 15:47:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9891dfd465a8bbde1cc6e893691a7cb0dde4299c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-02 16:00:41.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_09_2026_15_53_34\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"9794cc829fb2d20bfde22d739c8986af04b8c01c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-02 16:19:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/5/upstream-check/cinder-tempest-plugin-ontap-fcp/4652565/ : FAILURE in 27m 52s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/58d3ae3/ : SUCCESS in 30m 11s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/991057/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ab4c2c0/ : SUCCESS in 28m 51s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"8a9388b41477942e2f158d515e3485309ff78975","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 18:53:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-5-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"f6c2d43ad95043ee59c97cff07b362f36ebe3717","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 05:18:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-5-fc : SUCCESS in 1h 21s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-57-991057-5-iscsi : SUCCESS in 1h 4m 4s","accounts_in_message":[],"_revision_number":5},{"id":"04efd5ab52d3a6f9faa302ecb55e4bcee0ac4eed","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 14:09:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5196/.","accounts_in_message":[],"_revision_number":5},{"id":"0c16826f40ec99064f657b182bc1cd105ffbbbf7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-09 05:58:26.000000000","message":"Patch Set 5: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf33840ad527d43c55aa7c9a112d139f12242052","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-09 05:58:37.000000000","message":"Patch Set 5: -Review-Priority Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f32f68ef30581a5c297407e13165d679c83f502c","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-09 14:43:28.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"db3ba87b1808247c9c2442bfc4b1431d48874488","revisions":{"5f632a06f51cc678011163a281b3ae671df8276a":{"kind":"REWORK","_number":1,"created":"2026-06-02 10:23:34.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/57/991057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/991057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/991057/1"}}},"commit":{"parents":[{"commit":"850a4ad429d7531ab0d0f5099dd60c34a20a8793","subject":"Use urllib3 directly instead of requests.packages.urllib3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/850a4ad429d7531ab0d0f5099dd60c34a20a8793"}]}],"author":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:20:07.000000000","tz":-240},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:23:32.000000000","tz":-240},"subject":"[netapp-cinder] Added support for san active active capability","message":"[netapp-cinder] Added support for san active active capability\n\nChange-Id: Ie0a1dc78ad00d315e75e8ae2880943f39320c834\nSigned-off-by: sankethnshetty \u003csanketh.shetty@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f632a06f51cc678011163a281b3ae671df8276a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f632a06f51cc678011163a281b3ae671df8276a"}]},"branch":"refs/heads/master"},"b2efb28d0c548b29857872aaec807487ed4caaa7":{"kind":"REWORK","_number":2,"created":"2026-08-21 08:51:51.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/57/991057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/991057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/991057/2"}}},"commit":{"parents":[{"commit":"a25d8158d1006031cec16a81bb513ea9e2e0b6d2","subject":"Merge \"KIOXIA: stop global urllib3 warning suppression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a25d8158d1006031cec16a81bb513ea9e2e0b6d2"}]}],"author":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:20:07.000000000","tz":-240},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-08-21 08:51:49.000000000","tz":-240},"subject":"[netapp-cinder] Added support for san active active capability","message":"[netapp-cinder] Added support for san active active capability\n\nChange-Id: Ie0a1dc78ad00d315e75e8ae2880943f39320c834\nSigned-off-by: sankethnshetty \u003csanketh.shetty@netapp.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/999310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2efb28d0c548b29857872aaec807487ed4caaa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b2efb28d0c548b29857872aaec807487ed4caaa7"}]},"branch":"refs/heads/master"},"5c57bd89a320e12a5dfbf44963214fc71a6436fd":{"kind":"REWORK","_number":3,"created":"2026-08-24 11:18:43.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/57/991057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/991057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/991057/3"}}},"commit":{"parents":[{"commit":"a25d8158d1006031cec16a81bb513ea9e2e0b6d2","subject":"Merge \"KIOXIA: stop global urllib3 warning suppression\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a25d8158d1006031cec16a81bb513ea9e2e0b6d2"}]}],"author":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:20:07.000000000","tz":-240},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-08-24 11:18:33.000000000","tz":-240},"subject":"[netapp-cinder] Added support for san active active capability","message":"[netapp-cinder] Added support for san active active capability\n\nChange-Id: Ie0a1dc78ad00d315e75e8ae2880943f39320c834\nSigned-off-by: sankethnshetty \u003csanketh.shetty@netapp.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/999310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c57bd89a320e12a5dfbf44963214fc71a6436fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c57bd89a320e12a5dfbf44963214fc71a6436fd"}]},"branch":"refs/heads/master"},"4026c010e763dbe803d046ad170a7cd02900423a":{"kind":"REWORK","_number":4,"created":"2026-08-25 15:08:10.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/57/991057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/991057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/991057/4"}}},"commit":{"parents":[{"commit":"b7831e8de01aa78801a53a534ef75b88680a8722","subject":"Merge \"Quobyte: drop psutil requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7831e8de01aa78801a53a534ef75b88680a8722"}]}],"author":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:20:07.000000000","tz":-240},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-08-25 15:07:48.000000000","tz":-240},"subject":"[netapp-cinder] Added support for san active active capability","message":"[netapp-cinder] Added support for san active active capability\n\nnetapp_san_active_active is a new pool-level extra spec exposing the\nONTAP 9.19.1 SVM property `san_multipathing` to select\nActive-Active or Local-Active backends via volume types.\nIt\u0027s applied per-pool because san_multipathing is scoped to the SVM\nbacking that pool, and a cluster can host multiple SVMs with different\nsettings. Gated to ONTAP 9.19.1+ (older releases omit the key) and\nexcluded on ASAr2, where Active-Active is inherent.\nSince there\u0027s no ZAPI equivalent, the ZAPI client now lazily opens a\ncached RestNaServer just to query this field; the REST client queries\nit natively.\n\nChange-Id: Ie0a1dc78ad00d315e75e8ae2880943f39320c834\nSigned-off-by: sankethnshetty \u003csanketh.shetty@netapp.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/999310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4026c010e763dbe803d046ad170a7cd02900423a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4026c010e763dbe803d046ad170a7cd02900423a"}]},"branch":"refs/heads/master"},"db3ba87b1808247c9c2442bfc4b1431d48874488":{"kind":"REWORK","_number":5,"created":"2026-09-02 10:44:15.000000000","uploader":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"ref":"refs/changes/57/991057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/991057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/991057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/991057/5"}}},"commit":{"parents":[{"commit":"b475eafb370358e13999bbc1a8d6c7b4fe664500","subject":"Merge \"netapp: Fix clone_image() crash when ZAPI fallback is disabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b475eafb370358e13999bbc1a8d6c7b4fe664500"}]}],"author":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-06-02 10:20:07.000000000","tz":-240},"committer":{"name":"sankethnshetty","email":"sanketh.shetty@netapp.com","date":"2026-09-02 10:44:14.000000000","tz":-240},"subject":"[netapp-cinder] Added support for san active active capability","message":"[netapp-cinder] Added support for san active active capability\n\nnetapp_san_active_active is a new pool-level extra spec exposing the\nONTAP 9.19.1 SVM property `san_multipathing` to select\nActive-Active or Local-Active backends via volume types.\nIt\u0027s applied per-pool because san_multipathing is scoped to the SVM\nbacking that pool, and a cluster can host multiple SVMs with different\nsettings. Gated to ONTAP 9.19.1+ (older releases omit the key) and\nexcluded on ASAr2, where Active-Active is inherent.\nSince there\u0027s no ZAPI equivalent, the ZAPI client now lazily opens a\ncached RestNaServer just to query this field; the REST client queries\nit natively.\n\nChange-Id: Ie0a1dc78ad00d315e75e8ae2880943f39320c834\nSigned-off-by: sankethnshetty \u003csanketh.shetty@netapp.com\u003e\nDepends-on: https://review.opendev.org/c/openstack/cinder/+/999310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db3ba87b1808247c9c2442bfc4b1431d48874488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/db3ba87b1808247c9c2442bfc4b1431d48874488"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
