)]}'
{"id":"openstack%2Fcinder~992894","triplet_id":"openstack%2Fcinder~master~I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-13 22:31:12.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-08-13 22:31:12.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-13 22:31:12.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-06-16 03:28:22.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-13 22:31:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123","subject":"Bump hacking to 8.1.0","status":"MERGED","created":"2026-06-11 13:05:20.000000000","updated":"2026-08-14 07:46:27.000000000","submitted":"2026-08-13 22:31:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992894","meta_rev_id":"283b09adfd0ff751e5975f28bb8f5e40dab4b85b","_number":992894,"virtual_id_number":992894,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 22:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-08-13 13:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-07-07 19:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2026-07-29 14:01:47.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":1,"date":"2026-06-16 03:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":1,"date":"2026-06-17 09:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2026-06-16 08:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-08-13 13:28:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-11 13:08:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-06-11 14:02:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-11 15:08:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-11 17:37:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-11 19:54:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-12 03:56:39.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-06-12 06:53:19.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-15 17:43:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-15 18:11:27.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-15 18:17:45.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-15 18:25:29.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-15 18:30:49.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-16 08:16:09.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-06-17 09:38:37.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2026-06-23 00:22:52.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-23 17:44:16.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-07 19:52:01.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-07-29 14:01:47.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-08-13 13:28:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"3d9db8d59b1fadcf166546911c0eb28d07e77abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-11 13:05:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e61c9d7b1d62af4cc2ee3b3bdd9b0b4a284a0f2c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-11 13:08:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"783248cb4d6c875991efeca5ad1321bce53c18b2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-11 14:02:57.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123 .","accounts_in_message":[],"_revision_number":1},{"id":"bb936f50c998fe31fa8e1c2b04c5070e389217bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:08:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dca5541cf654209b4910b6d2326a912\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c0344f0b0644bc1a9ae1c063d49fc54 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da32276275b2476bbef4eb8a10584b3d : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81ce181ba16b455488ad742b7016ad3d : SUCCESS in 8m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6abde8a2bb834fb8a3e75620215803e6 : SUCCESS in 7m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05ff576a36ce438ab4e84c0fd56917bc : TIMED_OUT in 41m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e22d2888ebe4b0686decb89a647b458 : SUCCESS in 11m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4a8fe9db8b749b2b6107f7d109b8c50 : SUCCESS in 2m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d40ff411afbf43fab5dd23106cb394f0 : SUCCESS in 1h 34m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc8cc5bc879e4c44a4fa9cc4ba2c07dd : SUCCESS in 9m 53s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a74d5434259548a98aa8f09ece25de25 : SUCCESS in 2m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4a2b763e1b2e440783ae67f398ce6984 : SUCCESS in 9m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/732d16e7738e480996213123667e08b4 : SUCCESS in 4m 37s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cd3c6df462bc4d87bb13895069721938 : SUCCESS in 6m 01s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ff5a2dd97d974677b5890cf4fa0dd2bd : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9678c62ca6884fb5a4a0fc095d4ced01 : SUCCESS in 5m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5c57d23093c3422c9ae75cf7d9baca26 : FAILURE in 2h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"370e37429f37aba340967d44a8477aa9f211c48f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-11 15:23:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35672da2e77689aecd34863d11e9d0f133405a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 17:37:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa8cfcdc35da45868af4feaad2ef2034\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b70d06e5a6fe4000934d2d60b8fcc32c : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e639425ee1467ea3e0d113ce5a83f5 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ba5cf0ed1f64e9fa18006e4730a2169 : SUCCESS in 7m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71dc39e9c78b4393b5408951184f0bf9 : SUCCESS in 13m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b1b6429f3304f10a0ba59851940ca75 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f24054c2580140d98a50f2c6854c6bd7 : SUCCESS in 14m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca833ba479c246d89170cc3bd23f6f87 : SUCCESS in 3m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be27dd1d3519482d914b00b8595aa331 : SUCCESS in 53m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ebf3b9d70914040b996be0d6d30251c : SUCCESS in 9m 58s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e604f6b24fb743d3ac4781d9e1f27df7 : SUCCESS in 5m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e9d68c4012214b538e11bf303ede7efa : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d5957992717a4188a6e95a6f0ec35eb6 : SUCCESS in 8m 10s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8d3c264875ff4c5199a00023ca4c7465 : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1909264545aa4802b9780731afe7fe3d : SUCCESS in 30m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/91e68cb9eaad4769a552a66b2d72f6a0 : SUCCESS in 7m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3deee368fb024812b45c91423241c715 : SUCCESS in 2h 00m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aac8d8faf6c9cd5f1a2a7eb4a0604abe31e4c562","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-11 17:48:37.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_06_2026_17_40_53\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"bdc7a7238c6e4e59356b67bfac7c72d4da0193e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-11 19:54:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9972c1e8af264d629773079ce369c669 : SUCCESS in 54m 40s\n- cinder-storpool-tempest-iscsi-multipath build/63cfa7aaedf14ce79f4c73f59de35abe : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e3d96e716435f3c62fc2b7e32355fbd231c02301","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-12 03:56:39.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4456/.","accounts_in_message":[],"_revision_number":1},{"id":"a1750f4c53204a0928dba6b9483225f248c9fcb7","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-12 06:53:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"806460b3272651601d2f469c1b0502a804b6d131","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-12 11:35:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff85e79a4f430c47b1f96ee8aca52d22781f3d4e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-15 16:35:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb6c5febc009af522437e5344895827d4c700bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-15 17:39:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7266908b048802aba9ac01b6123a9ac3b9ded5fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-15 17:43:06.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/686a8d79b32c481e9e894341a4bd66a0\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d33f849b460476587acf195f1937e22 : RETRY_LIMIT in 28s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8a52e69eee0a4890ab9dfa1da348a08c : RETRY_LIMIT in 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe1a642f2c244b26843d77e7f43b083c : RETRY_LIMIT in 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f1a5a73e04f14f558b956771f3091a0a : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":2},{"id":"22d878a3c1bfea8fa479cb4cd783ea8f2a2b8d7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-15 18:11:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3c096626be8f402f9cb0f6d270ee0101 : FAILURE in 31m 32s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/992894\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/992894","accounts_in_message":[],"_revision_number":2},{"id":"f9ef14047fd48ab773495eb39caa7977c06c418f","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-15 18:17:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/6517220afc4a404fb7827625970c8d3f : SUCCESS in 38m 09s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/992894","accounts_in_message":[],"_revision_number":2},{"id":"33d7eb7e9820fd64c291ed9071635fcd2509d62c","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-15 18:25:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/6c5cea13a6af4faa82f3b91b3cb102f8 : SUCCESS in 45m 48s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/992894","accounts_in_message":[],"_revision_number":2},{"id":"c8e61a6186a307dba2fb523f770bbb76e99414e8","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-15 18:30:49.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/180 : SUCCESS in 51 min (This includes wait time)\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/992894","accounts_in_message":[],"_revision_number":2},{"id":"73cfb7f9dcc72f52345849ba9d6e06d159142c5b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-15 18:52:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123 .","accounts_in_message":[],"_revision_number":2},{"id":"459cc2cee77c5fb4f39f57f86df3420e22f50595","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-15 18:59:50.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/446 : SUCCESS in 1 hr 20 min (This includes wait time)\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/992894","accounts_in_message":[],"_revision_number":2},{"id":"5ac7137a36e7baccaae497c09ee94dc67f5cf191","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-15 20:29:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/dbe2348622cb4cc38d3810ad5bd60d36 : FAILURE in 50m 10s\n- cinder-storpool-tempest-iscsi-multipath build/356e7da025164756a4738ae65a843de5 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":2},{"id":"cd03ecdaa23ae60dacd09641da580a6771b31abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 21:13:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abb4a47aec314a379897f2baeb8655db\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5d7c01e3c26945dc92de76ec515830f7 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb07348a03194c9e80cffa34271f3093 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6bab9a82c1f428cb9b898d3c4b1566a : SUCCESS in 11m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2003c08b603e4a669172284076ff12dc : SUCCESS in 11m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a257cf9296f4a2ea662455791ef9f54 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4e3ef9f9be467bbb35f958c37fa827 : SUCCESS in 13m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd46675f53524eac860e10c4fdbc25d0 : SUCCESS in 4m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/3d0e0c3403394c7e9e83353ce7235d7d : SUCCESS in 49m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7cd0db31ae9446fe964598f847c77ceb : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c5f24fd097643968ed5aa856391ec2d : SUCCESS in 1h 21m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c79f89972e24ed89d68c18f10cb9de5 : SUCCESS in 38m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eb21ebc71ad498484954956d75878dd : SUCCESS in 9m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e25523fc853a4c93ba7dd168aee262db : SUCCESS in 22m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ef879c2b54ef4ba2a604777db83047fb : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f73e2f9059741328018e676d25ffb86 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/898b6da3e8b240f591cd899190cbed61 : SUCCESS in 6m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8294e3620f1a43a892f84f54518a719f : SUCCESS in 7m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/35704f9c724f45a5973b2f160fb4ca0e : SUCCESS in 9m 29s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7638c9ff1c834873a5df845a6c095ce4 : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a03fc086e6a42c28837a4b1378d9695 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8a57109ce5941919d8e767c0f7744b3 : SUCCESS in 1h 46m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4561de2f6efa4cbb9512f7efa0e514df : TIMED_OUT Host unreachable in 3h 28m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d2530268cc141fc99dd2712f9c444d6 : SUCCESS in 1h 45m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b439b8d6e6524b9283dff3de59f7de8a : SUCCESS in 1h 47m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0faec7e59cd048b28311ee369dea849b : SUCCESS in 41m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/047b8427cfc84c999fb761dbd765eb32 : SUCCESS in 1h 14m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9c2041b3fe54f76ad8a9287036746cf : SUCCESS in 1h 34m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a72e5406b1b145b7ac49aef0fe3f5e91 : SUCCESS in 46m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ba5bf0183e44efbaface4b68544dd88 : SUCCESS in 1h 57m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822c6d0baabc4a269a3788284c984233 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5c67ec1f51cb4d56571557353f0231d107da0e07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-15 21:24:12.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_06_2026_21_16_47\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"204de435bac358b5224805d468f6b9a3aed1a5bc","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-16 03:28:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e509a72fc0dc3d6aa14797a4d829b3f5137f23ec","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-16 08:16:09.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"76518360a727576c44c7b87fa7fb3b5e01c66170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 07:24:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5ee5e1a3f2487b9e96893a0290d3f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/445e90ae4f5d4b4cae471128a1f811ba : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/254bf568303443a6bbb51c7d78bc902e : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e09bc5c6fc264e09a789bfd9c39de4f4 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ccb6ade9af9f9335979a7318a62de7122728c03","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2026-06-17 09:38:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1e443c97de2a125b81a7a1cda69432c53a439724","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2026-06-17 09:38:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca3e724159cd466180797ab00e33556979613b14","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-06-17 15:09:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"711cd5ff271e44a0c798ad5c842d21d77611f935","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-23 00:22:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/94/992894/2/20260623032416/NEC-ISCSI : SUCCESS in 5h 32m 47s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/94/992894/2/20260623032416/NEC-FC : SUCCESS in 5h 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6e213e114443590b68214e66d4639864646acc2c","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-23 17:44:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-94-992894-2-fc : SUCCESS in 1h 5m 4s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-94-992894-2-iscsi : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c727e4a46f932979cacc7e3e842fcaeb113c813a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-07-07 19:52:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7fc80659ebc2f6130e826cf1ee513d616d234a3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-11 21:19:20.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4280/.","accounts_in_message":[],"_revision_number":1},{"id":"ffe033f761db740bcd9dfe97a465ecc59d4b2b62","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-07-29 14:01:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04e4cfa550561f1f0cf7a89e52c214b432b0afea","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-29 14:06:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2846485070b2d8f44bfdf238d2e726484930f4c1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-13 13:28:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29f9ec79c0eb93ee2cca10568181aea6b074da7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 13:29:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"671f78c0bac3d5b74a3f92bbfbc1dd5256075290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:45:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/079599a1701e45e9bf90830fd7b3af24\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/71401cee2f444b91b8c13f576d3f388f : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c169852ca7942dfb65be1894bc204ac : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55c55fd9d9284596a3b4a4670702891c : SUCCESS in 8m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d8b65b76c974a8aaa4e9c70b85f948d : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1c8ae15fd1463cb24d2e93d705a4dd : SUCCESS in 13m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/226810326f7b4b6f92f55296cb844850 : SUCCESS in 2m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/0cd4eb95108247aaa0f0521e5625ec85 : SUCCESS in 1h 02m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c681e359d5b451da674baea4124ca8f : SUCCESS in 1h 11m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c96cd1d1ade4fd0830fc7d858bb22e3 : SUCCESS in 1h 50m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20af77e96128463a84586f343b4ed229 : SUCCESS in 1h 00m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd948858ae774a37918c7010b0fd6dce : SUCCESS in 6m 13s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/108de0a646ce40c8bb0373beb123cd9e : SUCCESS in 1h 12m 01s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b12f55fe3e564cf4a69f0ed226e1e84c : SUCCESS in 1h 50m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc2cc01ad6294b239d1fc49c4a577b7b : FAILURE in 8m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d284b79d8a5d44173614c2a1082d2398208bc785","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 18:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86d9042aad64e2577fdabcf5bea45417f5188c8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:28:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/940a89505ccf45f0b0132dccf34c40e5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9162a41a42b74ccca106f13bdcec5b8a : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4b0ed1739a541d9877b1dd9579d7b84 : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/45b27f96b9fc4336bdcc7c33b7b64ec2 : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f10becfdc189d30d988bfd162bad3984ba58d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 20:59:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d41413420e848bc85a534d78a5c78aa\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c9fa0632e7b4e079cdb6e9fecf4768a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6dd71073624636a1a9b03a86ecc40c : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c1c6173aeef44d0b08074c2c6e3dbd7 : SUCCESS in 8m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e59a45b149e249f1ac5679fb8dc0be41 : SUCCESS in 7m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c65ce8da1640463c9fb052ca32f717e0 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87e2f3fa39a4dfdb8275d3870e38a41 : SUCCESS in 13m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa76183b7d2043e092878ad32e438d04 : SUCCESS in 4m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/0d5a99b5f60b439e919e8f3cd93599f3 : SUCCESS in 1h 03m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48aefd5c20534953830bd99b4f77d212 : SUCCESS in 53m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/317be23516fb4466afb760c04926781d : SUCCESS in 1h 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa7b6945212740c8b80a6e6627ff282c : SUCCESS in 1h 01m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7c8435b67a7465db438d83550e5be4f : SUCCESS in 8m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95180bd3034448c3ba2889ddf657802a : SUCCESS in 21m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/14b9dde9a3b740a3b87bf6b883cf6fb5 : SUCCESS in 2m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77c7ecc7f78a4b00b1b70a936ec86df7 : SUCCESS in 7m 07s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f84878f132bd479083b4d9de81da58c7 : FAILURE in 10m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9e359b2414334aa7aca1f2cd5ef56475 : SUCCESS in 5m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/126e5879483142fe95eb3241b1bc20db : SUCCESS in 5m 38s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c90535a40346401dbd5b6a7a1263885e : SUCCESS in 7m 47s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a05e0c17ea844235bb1bf94fca851e15 : SUCCESS in 34m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/590fb7e2d1194f99bf0c2f2b6616312d : SUCCESS in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9b94dea757d4dcfa514edc59e935705 : SUCCESS in 1h 37m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/def9fe72bdf04b728ae3269a585454d2 : FAILURE in 1h 01m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b222c59201b14148876dabdf64ca2f48 : SUCCESS in 1h 35m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b05ac08dcd0243b7aae7f346744fe183 : FAILURE in 2h 48m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6c9978afd2d3450ebe1532bce67fa220 : FAILURE in 2h 01m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/840ef62e87604c15a795f248e8d8eb71 : SUCCESS in 33m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/be90a25957d9486d92911a1a6b461ef9 : SUCCESS in 48m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/19c2a3c4e29e473e827b69ac81bcfb94 : SUCCESS in 1h 32m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1ed95f55cfdf4d0fa095ed58d75707c7 : SUCCESS in 46m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbc3db42f34e4b37b65d36ff29addfcc : SUCCESS in 1h 26m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53933542c5b44876bd4c0fe2582747db : SUCCESS in 52m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"c6b181ef658cdb92066973039bee07de56b55670","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 21:00:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1bdc51177a111eb84b2bf1027eac964cf8bfe7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 21:10:19.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_21_03_15\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"2c5c839b2b3d519686e25a2e3c46275f8808156a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:31:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508ab01ac23b4159996030988e61fc1b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84cb90d85abf4c62a008debd8bb86a66 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9618251bb1ac408b94ef7aeab4e9f43a : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7802c341726b49aca221317e39a76efd : SUCCESS in 10m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07e9d7ee22764c2abe2f5790ddfe4388 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f57c70f76d7491b9fe834f1fb937b3e : SUCCESS in 11m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2c620ed4a254497c883e38a38ea1884d : SUCCESS in 4m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b1683177bb3b45e18ffe3f00c9fb24b0 : SUCCESS in 1h 01m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82588cf1733b4ca599b9c68e889d0b49 : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b91756510384f4c8e83eeeccae5bcf2 : SUCCESS in 1h 25m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22d71371bb7a4788b9cc3d78def73e5f : SUCCESS in 38m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/926024ceb2574f8896f48b6ff4c4c5d3 : SUCCESS in 9m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e55c30b61b2413d93311e60914851e2 : SUCCESS in 47m 42s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ed8706432cfa4afda437c0876d37e90b : SUCCESS in 56m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef2255f2416d4bb1af5d1e94f027d2af : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f519947d8a3a8baf90dce1cb445b826956045fd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:31:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"94c34dfedb0aa0827b2cdab7f93673d8e0b47765","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:32:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae60e0caae443e69fdf945277968107\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e9f0fb91a3ce416f8d3b2ea52313b1ee : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9bd2cff92d44d5837592f07a9da950 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f303fa8c9878498e8a7427f761d68e4c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"0900f868fa59b426a0696e1b7237ac45f5965038","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-14 01:42:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/c28e93ef265e41ebb3125fc76fa2c039 : SUCCESS in 46m 59s\n- cinder-storpool-tempest-iscsi-multipath build/304334243a004b178a3f056233a78922 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":2},{"id":"283b09adfd0ff751e5975f28bb8f5e40dab4b85b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-14 07:46:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-94-992894-2-fc : FAILURE in 11m 34s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-94-992894-2-iscsi : SUCCESS in 1h 3m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72cef44b7185dd2ae4af590cddf3ae52d874413c","revisions":{"47b0ee5769acbbf6a4fd10fe2b58c099002bc87b":{"kind":"REWORK","_number":1,"created":"2026-06-11 13:05:20.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/94/992894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/94/992894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/94/992894/1"}}},"commit":{"parents":[{"commit":"7b775141853d809cc2f66c8d45710c3d3d6c2230","subject":"Merge \"Remove eventlet import from backup API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b775141853d809cc2f66c8d45710c3d3d6c2230"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-06-11 13:04:38.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-06-11 13:05:18.000000000","tz":-240},"subject":"Bump hacking to 8.1.0","message":"Bump hacking to 8.1.0\n\nChange-Id: I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47b0ee5769acbbf6a4fd10fe2b58c099002bc87b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/47b0ee5769acbbf6a4fd10fe2b58c099002bc87b"}]},"branch":"refs/heads/master"},"72cef44b7185dd2ae4af590cddf3ae52d874413c":{"kind":"REWORK","_number":2,"created":"2026-06-15 17:39:10.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/94/992894/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/94/992894/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/992894/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/94/992894/2"}}},"commit":{"parents":[{"commit":"cee90b0e496110c48166e0056a908dbb5356ab10","subject":"requirements: Fix duplicate \"packaging\" entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cee90b0e496110c48166e0056a908dbb5356ab10"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-06-11 13:04:38.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-06-15 17:38:54.000000000","tz":-240},"subject":"Bump hacking to 8.1.0","message":"Bump hacking to 8.1.0\n\nChange-Id: I9b8cf2bf7d4dba7460dd3af3d4a279143ed2e123\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72cef44b7185dd2ae4af590cddf3ae52d874413c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72cef44b7185dd2ae4af590cddf3ae52d874413c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
