)]}'
{"id":"openstack%2Fcinder~993363","triplet_id":"openstack%2Fcinder~master~Ib07d2cd496cdac490313b69e55b03093b012af9d","project":"openstack/cinder","branch":"master","topic":"review/993363","attention_set":{},"removed_from_attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-28 15:47:59.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"38371":{"account":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"38373":{"account":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-28 09:23:11.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-28 09:12:27.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-27 20:29:07.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"38479":{"account":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"38614":{"account":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-08-28 17:10:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib07d2cd496cdac490313b69e55b03093b012af9d","subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","status":"MERGED","created":"2026-06-15 14:01:40.000000000","updated":"2026-08-29 09:27:58.000000000","submitted":"2026-08-28 17:10:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":151,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993363-review/993363","meta_rev_id":"fd990694bb17b3f83f20c5c12c92b54bc1b37df7","_number":993363,"virtual_id_number":993363,"owner":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 17:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2026-08-28 15:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-28 09:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":1,"date":"2026-08-28 09:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2026-08-28 15:47:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38614,"name":"Sailaja Mada","display_name":"Sailaja Mada","email":"sailajam@netapp.com","username":"sailajam"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38479,"name":"Medha Choudhary","display_name":"Medha Choudhary","email":"medhaj06@gmail.com","username":"medhac1403","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:05:22.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-06-15 14:25:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-15 14:58:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-15 15:00:15.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-15 15:02:35.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-15 15:29:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-15 15:53:22.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-15 16:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 16:47:26.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-16 17:14:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-18 07:44:55.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-06-18 11:35:48.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-22 19:38:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-02 02:21:34.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-07 15:07:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-13 15:58:52.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-13 15:58:58.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-14 12:43:10.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-07-21 11:34:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-08-08 09:31:32.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-08-11 13:36:55.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-08-21 16:00:00.000000000","updated_by":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-08-21 16:35:19.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-21 17:18:52.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"7359a7d1cbb075d34fc8aa91819c19fb69d7dc96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-15 14:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4d6e64830860fdd810e33e03830a8374f533ee6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-15 14:02:28.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1cb0b5d021398c13730cd504df322afe33215b4e","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-15 14:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"1e10b87cd45c4f8cb1d8b2b1112af80989b06e03","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-15 14:05:32.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"5d8bced8c2234893ccf2319b55950f21d224e662","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-15 14:25:03.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a63036bd7dd04811aa6ba02b552ee0c7\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4646c8de7d5b42599982158553712d1e : FAILURE in 22m 53s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a8bab36fe0c447f59a2f4ce6905170b5 : FAILURE in 20m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c6448deb5a60475387645c1de0a8cac2 : FAILURE in 22m 01s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe971b39c31f4474b272bdce236b967d : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9bbcb910853beaa6331bfe5ff435f2c766403669","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-15 14:58:20.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":1},{"id":"893b6925785ca1a26276126cdd502fc19fc6ea40","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-15 15:00:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/88aea05bded74503b54f42ca87aff6b1 : FAILURE in 33m 25s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":1},{"id":"f3958c4e885c643ab5c9a188a45e8c3f0803055f","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-15 15:02:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/f0ab89aa90994fdfa6eb2046f190870b : SUCCESS in 38m 40s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993363","accounts_in_message":[],"_revision_number":1},{"id":"f30d4baadc2df9eebcfd96c3521e06769b692a2b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-15 15:29:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/f6c991ca494141d1b8b357cd6923d34b : SUCCESS in 39m 49s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363","accounts_in_message":[],"_revision_number":1},{"id":"9e1a551d7f8c2d8f21e76cb9fbd1515624e48c91","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-15 15:53:22.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/177 : SUCCESS in 1 hr 51 min (This includes wait time)\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":1},{"id":"f3a860d07d1323a774cc1fffbe745d699f423c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 16:23:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2be453e26254c8f831b4b515216c11d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85816e22c994416b96cc296b9da3ccdd : SUCCESS in 9m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b0072712ad94e3f926d3662a3d3aed7 : SUCCESS in 7m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7fa3dba318064840bacb3cd4822776aa : SUCCESS in 9m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5ff4a7355471488b9d52076cc86e1c59 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58344c52d3b6484bbdf4281ccf4a73db : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9950d905ad0f47999617dbdd96dd02c9 : SUCCESS in 1h 07m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66258811a57c4e35935aefc895bb51c3 : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f2447c201644663a5d9ffdfa5b48d59 : SUCCESS in 1h 43m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21f08698cead4f1bb5cd8dec6feef857 : FAILURE in 32m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1093d549910416abb57b3146833bfa2 : SUCCESS in 8m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28947ac61c764b83a08d8c6c9ac84d8b : SUCCESS in 28m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0a5f687d24154ba3b9cd4ec3a38923f1 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7884cc0abc9c47a6aaab3b6ac04b4be3 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7a90f991607a4eeabe133e61fd86a04b : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b00d5c65f54f477a9c22d505cc47d316 : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1eda2b1608524c15ae0e04c4d0a628a7 : SUCCESS in 10m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/250c8de83a9c495d8a2f390d38d37915 : SUCCESS in 41m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d499af4b767462c89421d75e959854b : SUCCESS in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/17e57b9cbe0f4317a0685c205287022c : SUCCESS in 1h 56m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c54ec6c67dd42aea4881e233ade9d37 : FAILURE in 1h 48m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e79b56e65cad4bc692cbc09139c23cd6 : SUCCESS in 2h 13m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/840e446531f24671a6573dbbf8763df8 : SUCCESS in 1h 38m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3d915f42dec041d0af6edbba880288b4 : SUCCESS in 42m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b7b3a78284e445de8fdf83f94e4008ca : SUCCESS in 1h 20m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a37e3ac64dc94ebe9127249caaa542e1 : SUCCESS in 1h 28m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8a44a25a11f14af49a1506d022f7d995 : SUCCESS in 50m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2295462eb0d248dda1103b555f057e45 : SUCCESS in 1h 30m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b46759a534b84287ba0a15ebed050469 : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a0c470ce0fb4e158f82a6262e214ff9d061b13c4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-15 16:30:09.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/443 : SUCCESS in 2 hr 28 min (This includes wait time)\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":1},{"id":"b576cc62093bf6bdf56b4df22618a839f995974b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-15 16:47:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/a1f9f48de38e4a8b88a4fab74de6eae3 : FAILURE in 51m 52s\n- cinder-storpool-tempest-iscsi-multipath build/77c18044a29944b598488c3090143189 : FAILURE in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"282abc3ea4a2547a7a88358009a569a9560c2da4","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-16 04:49:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97367fe3ed272284c922ccbf9e9ced8e9beb3b1e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-16 05:13:12.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6f12b11c0a5b4b8a8ac20c77f38b6c6a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/70617e4e92be47b2ae751e481e52dd01 : FAILURE in 20m 32s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4b9a20864aaf485fb348e6269a58f347 : FAILURE in 21m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13c8b973ad474bff899a67ef66dfeb00 : FAILURE in 23m 23s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/578dd5ed3dac44e588d6ecb1d0fd336a : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"62be307e025473291c15e9d4084440bc9a72be18","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-16 05:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/06539e1c166a444da04844fc570f1943 : SUCCESS in 48m 44s\n- cinder-storpool-tempest-iscsi-multipath build/057d8d9bdee949348d8cf4d09d561d35 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fba37dea77bebb40d4a773f76b2d01e73881a5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 08:41:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d67b2f0f56464b11ac796cd1dba1934a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ce642877ad4994aa563543e689e9ba : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/637a5151d8514a44af55813a7010fefc : SUCCESS in 12m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afd0171c99a64402bf9fe6142b69557e : SUCCESS in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71c04d86582b4f7eb9ac89dd84027ba8 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc6fc3f88004a8887ac29886c74fef2 : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/911179d1d54a464fab2293af6b6831ad : SUCCESS in 51m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40b24587d8ca4b23ac5920cc50fbe37e : FAILURE in 17m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d9c116fbd574b3eaab656decfd863f2 : SUCCESS in 1h 43m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/786afde0622140fa99c95d5e41271a06 : SUCCESS in 1h 07m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d89075f966a64e969f2e17242e221a65 : SUCCESS in 5m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5f6a7a9f20274f1bae44348163b4e297 : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/76b4e60cd63140f0b7f8277148067510 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd60bba777fd4ea088d620f477c7bbd6 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7c4fdaf78a604319a88954757aa805c9 : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/91dae9ccd38f4786a9b86fa9abcbf495 : SUCCESS in 4m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4a1adcf0592e4dc7909bf549839aebad : SUCCESS in 9m 31s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ebcb15b9efcc499ca519b9ffbf72df7a : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcd6caed31204086b3eb2a6e3cc14c4d : SUCCESS in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d3b092144924f43a44078c554c42849 : SUCCESS in 2h 39m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fbd5683a5f62465f905ca2f965354928 : TIMED_OUT Host unreachable in 3h 45m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7a4147def0742c89b3e5a6391e2099a : SUCCESS in 2h 03m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f2264874b1b342f6b2e1b12453268f92 : SUCCESS in 2h 00m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/623f8364f84f4d6299eab402b5dc4ab6 : SUCCESS in 24m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/89379c0796d14d57b733c58a22d670dd : SUCCESS in 1h 10m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d0756dd8d534f8c8def1c43ee18a40b : SUCCESS in 1h 55m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7fd3f83577ef4196b0247ed84fb97a77 : SUCCESS in 49m 16s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/33496ecb5e414c6981c1ab33a5cabb77 : SUCCESS in 2h 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1be7d126f5ea49b496da1ec0e626806a : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1d6f2b2e2de16c961857ac8403a08cfdda6ed72d","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-16 13:41:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3386136a3a769d882f27f241cfa20a00719e7ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-16 14:05:34.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/87b3aed10513455ba2ca7b9fb5c4a53e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1171f8dd8894bbc94b4aa5787e09c7b : FAILURE in 22m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a0345da8c1c64008beba0b0bcfdd4260 : FAILURE in 20m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/411aa497d9614ea588b2c6bda1eff0e9 : FAILURE in 23m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/114f3e8021d04e37ac0b475ca885ab18 : FAILURE in 21m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f00f85db19ad6fca040c81ad759edd6d88be2493","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-16 15:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8f033506e4c34dbda40e5f32938966bf : FAILURE in 57m 23s\n- cinder-storpool-tempest-iscsi-multipath build/fa556123c27045e3a62aab3b685d4c12 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8013f8dfdbaca584844f13a9eedb2fd3c6862329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a6e4146b63e4584beeabc47bd32d25a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceee015fb755462c9b8ba82a87ca56ff : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7a35ec7a11840c09da1981aeb444df0 : SUCCESS in 11m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55e77a58a8fc4035ad1b71117397f09a : SUCCESS in 7m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0924eb79c85a48088222d099a58c7c7a : FAILURE in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a24bd928dd40fdab1c39ac43550b54 : SUCCESS in 15m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/8dff4f8d409d45e08dcb5f5bb7adcc24 : SUCCESS in 53m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d43cfca1cce54720a7da85fb9ffd2329 : SUCCESS in 35m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/31d8db2a32ab4e6194925ddef6157cd7 : SUCCESS in 1h 40m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bd54670e3e64e2aacb98fa0654b383f : SUCCESS in 1h 09m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3833ac0179684fe2a4e17c9a5c542238 : SUCCESS in 9m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8287fffa58f7409b8d48486b88811140 : SUCCESS in 34m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e5b6c6f792ce4b5fb703606e5f8cef80 : SUCCESS in 6m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20e322a311cd40a18f2719075d097f15 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1ee7dd3be3a54da8a467123017d7c491 : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8beb4a670d6b438aa391b7ae0c8bed88 : SUCCESS in 6m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6073e7636d4b4268ab9775daf9b7069d : SUCCESS in 6m 29s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2c5dd8687db45bc8e569b0d6e1f14b0 : SUCCESS in 40m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2ec4b9786054b67bc593a19e8c788ae : SUCCESS in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36a9cfb0e00c4fd8add55425725d1f5f : SUCCESS in 2h 00m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f8f2075ad8b746ed975d53c5e8f15c50 : FAILURE in 2h 19m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e62e6e20a54943b6a8f6b64510a578f5 : SUCCESS in 2h 03m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e8339cc044864df198f3bb2c8d4c3be9 : SUCCESS in 1h 53m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/da6d6a75744145608fdc331906a3455a : SUCCESS in 37m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d832fc342081429d81f120bd362c3410 : SUCCESS in 52m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/44adce9f39c44301bc9654ffb417fe43 : SUCCESS in 1h 27m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f8dbecb20bc452eb94c84358f83cf62 : SUCCESS in 51m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58e1a7c6e6d54bcfaf303549b6b942f1 : SUCCESS in 1h 30m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37222d0aa88d492f81921967c758810b : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6dd9684910b6dafa5255cec34f26b2fce42fa85a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-16 17:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_06_2026_17_17_52\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"7e84e679536d829a094a68a94c0a91567f7d18ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 06:15:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f4fd391e11461cb2f40d5f1c08ef91\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/36f1cab531ba47deacd4631409cc1d3f : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c846e0f1b3b43c8aa5a0c505cbbc72c : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c7d7d3bb55541cd9d67f8503fbbcfd2 : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"499f37b92635ec983426a6b8d62274106e2b5a35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-17 10:09:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4b01405615827ea455942783d0fa89b85456672f","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-17 10:12:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd902f068d918d8f91b9c36170b3f62fc687a00d","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-17 10:13:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4befd17731cfc3757b46f539b414e64284fb382","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-17 10:44:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/700d0f16a6914fdc885db2b6c26a8ae0 : FAILURE in 23m 14s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"3b403c6b666cb8c97a3c3e688c556a05c5c225cd","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-17 10:49:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/c13d644124a646b6af264164e7cfc456 : SUCCESS in 37m 18s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"dfd7ca4fce140e4f8cb35fd2e33e39249bb831cf","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 10:52:58.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/481 : SUCCESS in 43 min (This includes wait time)\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"744adf5c12793546ac0ffd8bf0bfb70361bfec42","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-17 11:06:27.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":2},{"id":"20bdc556f138e9bf589e5ea01bda66d87a9952cf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-17 11:25:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/3cd4a4a88b3d411bb12eecee92985c80 : SUCCESS in 39m 46s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"2247bc88a38d41ee3fb585d921b7da72bc81908b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-17 11:45:02.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/869c3109c29247bc82ff57dd3b6bbb7d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28b7552a505846f9919977c59e96dfa2 : FAILURE in 39m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4265c1deefe44abb82fcab5df1ae4bd1 : FAILURE in 42m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2cf7f571db7f418787576c417df01984 : FAILURE in 1h 35m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7c477d8b2224387a2572264296b578b : FAILURE in 32m 22s","accounts_in_message":[],"_revision_number":2},{"id":"76b05a81551142df2e66f57d5b340d406fe4988f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 11:52:58.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/132 : SUCCESS in 1 hr 43 min (This includes wait time)\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"adb5ea087aca7f65c2e7a7acec658b2b66fca7fe","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-17 11:58:38.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/218 : SUCCESS in 1 hr 49 min (This includes wait time)\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":2},{"id":"59d5ffea35ce4588fddc03dc42bae14a46e7d56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 12:51:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a77374986d6f4bdaa6999188e71b44f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e5da6d66e546ccb5a2de5fd7ff6cda : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad03378b58d7459b965806880b7abb8a : SUCCESS in 12m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe71725f64de47e6b182d5b29abbf7ff : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f157666ea7a43cf8c68f2511df8ee33 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecd0670b9c6441b98bd11c0af4f80e5 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c19becf39ced4de5b161f19b43cacf02 : SUCCESS in 31m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac68fb80ca1a4b44bedac63b30869278 : SUCCESS in 1h 08m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09ad87446f794ba0b46618e2bb404447 : SUCCESS in 1h 17m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c8735fe39d0486eab3c42077681be59 : SUCCESS in 48m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0f005f1dfbc4707aee3c4ffa6c6f2fe : SUCCESS in 7m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb06f15962a4465989466ef5085fcc6a : TIMED_OUT in 41m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e92d16390364db7b2aec53c9367c051 : SUCCESS in 4m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3baefd816744366b95bef8d1f2b706b : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ffc1928f98524a7189a68f03cca0d9b2 : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2b27b7530b7e4acda31890b9ddd1f446 : SUCCESS in 6m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/090bdc0173de4e43a5003b8586ee8f84 : SUCCESS in 9m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2245417b3c4240c7b120abdc799524c5 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/709571ee5cd840af8d0f5ac42950622d : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f6412364c6a44a3969201d3f4088bc0 : SUCCESS in 1h 01m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/68f8ac48ee7d4584be19da14d391420e : FAILURE in 2h 39m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cd932ba4324b425dbfda673fd30792a0 : SUCCESS in 2h 18m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/82ce6a24a53540c0873d672cd281f418 : SUCCESS in 1h 54m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e0cf03b497644c30874d050398d16279 : SUCCESS in 26m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8a7570ed7fd4fd78e693c4b40ae2ebd : SUCCESS in 46m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ae03f11021b34a389eb7dd6001249256 : SUCCESS in 1h 38m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88e69505d1c747cba05a7a1040ab2190 : SUCCESS in 40m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fda4c72b15241c4a1421a9cb561387e : SUCCESS in 1h 05m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb24ba1c13334323a88f26335cfe3300 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":2},{"id":"af9096b6ebcbac10bdd5b7927804cf70a04d45dc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-17 13:10:22.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_06_2026_13_02_47\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"c67b73c08a54e9924d62c127d853a15ae4a833ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-17 15:25:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5f3d7933368845bf8b955a08e04b1688 : FAILURE in 55m 36s\n- cinder-storpool-tempest-iscsi-multipath build/379e202047354811ad524bfebac8a7df : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f1b0192f74db0cb09a1c4288acd82cecc3d3d19c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 20:51:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fccd2a5a1d08441ebf4f8b9a27b8b50c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e095134dd81f4a48876664980d5fb9a2 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72b5da31d3aa4de89db4fb1aecb9e710 : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a0acad9f89c745a6ab96d1786b22369b : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dec3301a429fadaaff89e718684a17c90e570b68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-18 06:07:19.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2efd4ac4a7ed87270284425af4b3987d1892ed8b","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-18 06:08:26.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"629355cf39da465ecb0c210cb73cbc11e7abd03c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 06:26:18.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/491 : FAILURE in 0h 18m 35s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"c59ed791f28f42d6c763044dd6f2e590b64c8fbd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 06:53:22.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/227 : SUCCESS in 0h 45m 38s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"f797d60bbdc7c2e182ce9e140dd31ab1bc129b19","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 06:57:37.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/141 : SUCCESS in 0h 41m 1s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"00e014b2a9ed0f0ddb733d14909958220b3d165b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 07:05:33.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":3},{"id":"49163abf7b73481f01dda50005c5a53087c339c2","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-18 07:06:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-fc-driver build/7e18d8e213a54ca88623180d9aec8bcc : SUCCESS in 38m 02s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"c095455bb4ce6219c4b70316f06a3bd588847725","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-18 07:12:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7869eac55e66dd4490dd1d7d9420e738974ce86","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-18 07:12:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/625a50d7ab0142d2be571f95a04f87fa : SUCCESS in 39m 36s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"241a368a9a19239684e3d39d330f6457040564b2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-18 07:16:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/b257c21bda0847e687f12f635df466b9 : SUCCESS in 47m 13s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"8ba08b35702227c35776ab37dcf44e31852bd906","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-18 07:30:18.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/15cd990a09314cdfb8c4ee4afa7e146c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/646a5b7d88a845e7acfc7282d6b95109 : FAILURE in 34m 20s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40aff17a23504fccbcac62b26ce4f4f2 : FAILURE in 36m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28a303a1ea844744a98d14d79977c69d : FAILURE in 1h 22m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/84184e84891c46868f6570ae922cbae6 : FAILURE in 34m 18s","accounts_in_message":[],"_revision_number":3},{"id":"3e575ee527f1ae9bc9ef3f885c25d50b6ad14dee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-18 07:44:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/3/upstream-check/cinder-tempest-plugin-ontap-fcp/92e006c/ : FAILURE in 29m 49s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/0e2793c/ : FAILURE in 15m 09s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d10710f/ : FAILURE in 15m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"024e9db8cbca86338bf2c202bbaa77d347372aad","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 07:48:55.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/494 : FAILURE in 0h 29m 54s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"f6633c58b104bd94128bb232b3578825a813bcd0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-18 07:56:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/167f8d910d9a4600a4997de87ec108a9 : FAILURE in 51m 40s\n- cinder-storpool-tempest-iscsi-multipath build/73f61e1cb111450b9badf7ec5bed4241 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":3},{"id":"f916115732df6ad975b1622a280d1c6396e97143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 08:21:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3096d3a2dd4f498ebaa1a18658af0298\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a679344306cf4cf8b701053f19d29343 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bf07503f93c4966be12b1fc6d06232b : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb117a75101f4457b6a53d21b6a82332 : SUCCESS in 9m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efabb8f260b74b7997fb4192dd715ef2 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354c6d8b511d4c1b9a7025787a768cc9 : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/51fb7b31a31e43e7bd593c72cfa7bcf2 : SUCCESS in 1h 04m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/593b29f7dc9845108af9180350f3371d : SUCCESS in 36m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/009ba0c2cc3949cb80193eecadaecf68 : SUCCESS in 1h 33m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c1c214fa318486eabe3739a75c84589 : SUCCESS in 1h 02m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c53b3549a54a728423a155b6216e57 : SUCCESS in 9m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95e04afdd7c5490cbd12c929e54ff4f2 : SUCCESS in 20m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20de8acf866044638ba0cf25364bdfb5 : SUCCESS in 4m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48376c6c976d46b4a93d09c233f9de8c : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f79fb0e01c12494c95e205103b6d89b0 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1190cd99b163463e902e04232acffaea : SUCCESS in 5m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9a5f9922bc544030b4852b7bd6b29331 : SUCCESS in 8m 40s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4df580c7c889477b9d653ae50d7cd305 : SUCCESS in 30m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c53dce54feb437c9833bb5fd8b2a106 : SUCCESS in 3m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee1131cc079b4d878c883123f014b49f : SUCCESS in 1h 44m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e91ee27dd43848a4a3392e64ebc71103 : FAILURE in 1h 23m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1cfb21d1a000433c840f245e7e529faa : SUCCESS in 2h 03m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/da68cb1336f94dab863b8358f7988b4c : SUCCESS in 2h 08m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/77ee7a3d86dd4e6889f782089389a14e : SUCCESS in 26m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d48bf57b73042aba6c90e20c62a81e3 : SUCCESS in 36m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f79e1739569e43bab1fe613b3c168a28 : SUCCESS in 1h 54m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d8405de5300d42838cf3dcc68cb4f40c : SUCCESS in 45m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4bc2ff5ceff4732a3e3405f620593cd : SUCCESS in 1h 40m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3cac23f9f4f48d099f8c58286cb6de3 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":3},{"id":"89200de0a26809c80d128a10d050d91d0f76d810","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-18 08:33:01.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2026_08_25_11\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"d8d63da86cb0adaf8748fcce734877eb91a55c70","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 08:44:56.000000000","message":"Patch Set 3:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/495 : SUCCESS in 0h 40m 47s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":3},{"id":"d11eed6784e4fcea77ada59bc9c0b98d1d888bb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:52:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd985ef4620c4747a839838b968c7652\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc446ab0c18f45088801ff8534aae063 : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d950615fffe4f68b6052e82945547fc : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b24b654ab2ef4997a41e7a360e1cbf22 : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9d79f148ecc59c409efec0c969a0e2a061beaec","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-18 11:35:48.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-2-fc : SUCCESS in 58m 47s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-2-iscsi : SUCCESS in 1h 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1c366aa7e1be6b35d92db111d212762b67c0856b","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-18 14:59:51.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ad4fce5934f19c28f2126bf49b091b19011fc774","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-19 00:17:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-3-fc : SUCCESS in 1h 10m 16s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-3-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"324fe088448ac97e8ecaf071b3a289257c4e2d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-22 14:16:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"62e4c3e78e9d949cfdbb80b1089a66319bc35f42","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 14:56:29.000000000","message":"Patch Set 4:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/267 : SUCCESS in 0h 39m 17s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"2e8725092d109f3a3d7359129befd25d913c979d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-22 14:57:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/560500e6f9dc453981cd500df9d646a4 : SUCCESS in 39m 47s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"f04c6d10cda62a2a363c0017ee15f3cef083bc78","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-22 14:57:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/4fdc4ebca1844c31988730311821b4fe : FAILURE in 24m 04s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"0784f1ef4562c88a11d1fe140a5647957c4a2c83","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-22 15:01:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-fc-driver build/3632ee37628a449c99ee38e8ba5cbe9b : SUCCESS in 44m 05s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"86ac61cb8c577e0e406bec2206d0af3a38c69390","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-22 15:13:50.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":4},{"id":"9c957fd72f60a88c819f992b197df130576b754c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 15:26:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/358ca19b9e6d412e923b678e457b6b6e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eafdf01772684f729ff7837a7d038205 : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/163be89bc9514327a9058cca02117910 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e47816d7c5cf436dba22f64197d7276e : SUCCESS in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6cd2c7af76e73e9a054cc15e2731fb2db2c86f57","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-22 15:51:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/452875bbd56e4225b8ed62e79dfc6f0f : RETRY_LIMIT in 8m 17s\n- cinder-storpool-tempest-iscsi-multipath build/d39fc5dfe2734f82a55b1cfbdc90b176 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":4},{"id":"3b9d5582615ea64495704562c17c728acd7e3ff1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-22 15:58:54.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8f49d8217a3e48edab54288faa2a3a1f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2537bb1d27a04c1eadf62c070b0bb57d : FAILURE in 39m 12s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c75042d5702b4a46ab797e73702167c9 : FAILURE in 38m 41s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd1ff0321c774e2299624ab4460e9ca2 : FAILURE in 1h 41m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/785e7b51e16347b4b692708a7a494907 : FAILURE in 40m 29s","accounts_in_message":[],"_revision_number":4},{"id":"f335cda7d3dad8beb49461110ecd89c4ead376e3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 16:15:43.000000000","message":"Patch Set 4:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/181 : FAILURE in 0h 13m 6s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"fb9e2c8b008c45137399b3ae0bb38ecae9af855c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-22 16:41:58.000000000","message":"Patch Set 4:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/542 : SUCCESS in 0h 40m 5s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":4},{"id":"87ad3cee2068c5ad43c47d629ab403e0413f54aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 17:01:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4872d6adf8f54bc881debbc74d680129\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f7c777674a42bebe277a2e65e752e5 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/912f19754dd346029079e0c6170cebbd : SUCCESS in 8m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a92332316e04223bae0a2b0cfe2ff41 : SUCCESS in 9m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2e8ac74da4b4fa3aa2d50c98bf7b306 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce330768e58a49dd83e7472f56c254b7 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/1be6e6ab1c5d4ca59500c7f3e4615df4 : SUCCESS in 1h 01m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77e2accfbd314d1a9faad14ec2875660 : SUCCESS in 1h 01m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69524655c0a94f12852fc8828ce146f3 : SUCCESS in 1h 52m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b40af93702c4011805aa16ca8306890 : SUCCESS in 1h 05m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7c42c5d34a64b87bf83e9bc89ab8b89 : SUCCESS in 6m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4d18b7c613de49af8306c9c3979e281b : SUCCESS in 23m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e0e59129498460ab3ebcc723add46d8 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/424ad31d7e564122bf933b6b9b567e8f : SUCCESS in 3m 23s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/2d6355015aae4cca96ab18f7635407a3 : FAILURE in 9m 03s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/000cd4bd77e64801b4f8f6d2dfa30285 : SUCCESS in 6m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8ad7a54c627e48d69cf6fe927da6b185 : SUCCESS in 5m 08s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bf0b72f95a0a40c0ac8cc41abe4a90bf : SUCCESS in 7m 16s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e910b4f4d54845dd846e3be84639edcc : SUCCESS in 38m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f314bd86c85b433e8061032c7cacf80f : SUCCESS in 3m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af6da39f3b434646afd18450ccea7cb3 : SUCCESS in 1h 58m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bc80e453462a427f82ec7f10a8e1b91d : FAILURE in 2h 37m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6cd4bc208218450192c7d11df263064a : SUCCESS in 1h 46m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4047ce0515a3466ab89cbf0d7b572c30 : SUCCESS in 2h 24m 09s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ad43befbcc354893a5fb3c0dd8aeae55 : SUCCESS in 39m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc033aa247a14c349fce59e58e8e743f : SUCCESS in 43m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9571d5050cbd42c8b4ad1660d15156af : SUCCESS in 1h 25m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c1844b5f5e2c4636bc9eff8972df4e90 : SUCCESS in 44m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4f01c12e77b4c3399e032ad4c9b9a9d : SUCCESS in 1h 41m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47b8149494574289b9627f0b1278aa6d : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":4},{"id":"5020faa2ddc8876839bad88c449ee3e03513b61f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-22 17:12:16.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_06_2026_17_04_38\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"0d7115ec885f658bb472bf25f76db533d7e3611c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-22 19:38:55.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4474/.","accounts_in_message":[],"_revision_number":4},{"id":"02902d814c3c95c10bc525b6b70be932dc5ac974","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-23 01:43:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-1-fc : SUCCESS in 1h 13m 52s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"63541e785f74004fcdbc9c687fbdb7fe2fbb2982","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-23 07:17:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-4-fc : SUCCESS in 1h 29m 58s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-4-iscsi : FAILURE in 1h 26s","accounts_in_message":[],"_revision_number":4},{"id":"3c1197e8326b1aa472dcbfcad4e5cd2fa4687ca1","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-24 10:38:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"385ec3667399895397ae29a2d11bac11276dbded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-24 17:50:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"6be74a7751adc8faa5862d94659cc2907e6a15b5","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-24 18:22:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/83e27df88b6c415d9ee659ccc60b1997 : FAILURE in 30m 48s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"969bfed3c343444b7301398727b5956ab7ca9810","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-24 18:31:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/8b30de43cdf647a98cda8fe9c5c0aa27 : SUCCESS in 40m 41s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"c2571cb39ebc44dfb6152b78310ea10a57df1f0e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-24 18:34:39.000000000","message":"Patch Set 5:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/222 : SUCCESS in 0h 43m 36s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"240965567ad769c9f0b1177950f5e675be3572c2","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-24 18:34:50.000000000","message":"Patch Set 5:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/586 : SUCCESS in 0h 43m 47s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"ea8e620986c2a2ec19dbfadedca39c6fb661e8a6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-24 18:35:06.000000000","message":"Patch Set 5:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/313 : SUCCESS in 0h 44m 6s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"a0b30905862bbe6894ad08488a7e1ebc70dacfd2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-24 18:50:59.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":5},{"id":"e40ad5a5f071ef11f4b009942e9d4afab2714f58","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-24 18:54:48.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/dc6e161d0c144769a746b4a32460546c : SUCCESS in 1h 02m 18s\n- cinder-storpool-tempest-iscsi-multipath build/115b4be83fb64dc1804ef2bc526c9bd4 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":5},{"id":"70e82be648fd887eaffaca9a0971116f57736e47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 19:47:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c460a0bc71284bb28e8a6c9654a3a947\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12e96b883d274f3a92bca6c1742dfa99 : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05d8f71e23bd458aa5d8514b3a2b723d : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9d198fa0d8fc46e29f2e99215487174f : SUCCESS in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3589a4d0ec7fde95d9add4f29b6c634713661807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 20:24:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d192a975e54f3e83877ffd81f64062\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc794844f604253b3c1ac1bf5f843f5 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9610687c19374bb88fd9c0231bca3124 : SUCCESS in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da366de0c1c84c9f9993fd3b6891ef55 : SUCCESS in 7m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dffeaa47f8d3487083105c77eb45d345 : TIMED_OUT in 40m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098f9cf8fb1c4398910bafc5d73a1b89 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/d70c6c9c3cab4ce88f25949a80d8b579 : SUCCESS in 34m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3d09382f68848faa1a4601c568f3c5a : SUCCESS in 33m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0103f36ebc07496285605f28d8bb5613 : SUCCESS in 1h 26m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bed87f36925f4972be09e15ae337adaa : SUCCESS in 1h 03m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f364aded9f640c2b238837d2f02d44c : SUCCESS in 5m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c6e7f89b25284103a68ddba810193f73 : FAILURE in 23m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0bbbbed4a5d34dce93ee02090b8e3071 : SUCCESS in 5m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aafc7c6e99c94489a5bef70a52969ad0 : SUCCESS in 3m 47s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a23b3543e7b149499178187dbd550065 : FAILURE in 12m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6aa6e92834a942ad8a3777d286a6c47c : SUCCESS in 6m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1bacdb183f2847d98b501da1f98b25d2 : SUCCESS in 5m 48s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8d295a0672c841718a1808273a3cd9e9 : SUCCESS in 7m 14s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/97aea37599184bc1b953e479ca88ae35 : SUCCESS in 33m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2cee94c3a8524759813de119ca6b85f7 : SUCCESS in 4m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/65fda30c07614ecb98ca5edc3090b7e6 : SUCCESS in 1h 04m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a51b58269f24f4bb86d4f8b1fd34b9c : FAILURE in 2h 08m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef061dba323948319d0fb1b078f9bfc1 : SUCCESS in 2h 04m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8888d63cdffc452cb07397fb575c8463 : FAILURE in 2h 30m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/988bb5de63994d68937f36e67668fb42 : SUCCESS in 2h 01m 38s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/bcb38e7ce6f14ea684da3b978eb0b617 : SUCCESS in 21m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fcb47a828c6748a586001a81c546c6c4 : SUCCESS in 39m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9739c362ca35449d9622f5c4d872a895 : SUCCESS in 1h 18m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e5ce2740673d4e7c892b8adf0aef4d66 : SUCCESS in 45m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6556d13c0b4144c28135626b36a37440 : SUCCESS in 56m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60e621e040a04ff2adb490d81ce4a2a5 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":5},{"id":"1416cb50b83be6110815c395fde125aa3a146b3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-24 20:35:41.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_06_2026_20_27_55\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"299bbbe3e2e644de2ea4c0040198ba22bbf2b74b","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-24 20:51:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/98192bbe6d0946fa8970b3c919437f41 : TIMED_OUT in 3h 00m 42s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993363","accounts_in_message":[],"_revision_number":5},{"id":"35acd3218489975470d37b3188aaeb4b84338dc2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-24 22:49:32.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4514/.","accounts_in_message":[],"_revision_number":5},{"id":"b05e44fd789b6465d387d8a03988dec890f22e2c","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-06-25 05:48:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4b0b28bd22f69909a4d3332df0e5dedb550464e7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-25 06:20:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/5/upstream-check/cinder-tempest-plugin-ontap-fcp/ea60888/ : FAILURE in 27m 35s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/e5807b0/ : SUCCESS in 28m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c19849c/ : SUCCESS in 28m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"47604942cb509077ebe027c2b616fb23f9698c2d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-25 08:47:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-5-fc : SUCCESS in 1h 4m 46s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-5-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"1b032ff0f04d360837984d94420f6298f7c27114","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-06-25 15:31:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"20fbbd5b7ef30f57418679d9b3c55925720d2947","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-02 02:21:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/993363/5/20260702052306/NEC-ISCSI : SUCCESS in 5h 30m 27s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/63/993363/5/20260702052306/NEC-FC : SUCCESS in 5h 58m 25s","accounts_in_message":[],"_revision_number":5},{"id":"162eeb0e360406c09360519b97beed818b70dd58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-07 12:33:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6314f9794d6343e7542d3c829800eb738c194429","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-07 13:09:26.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8288e51d4c1543f78a1ffef97ab1f313/","accounts_in_message":[],"_revision_number":6},{"id":"aa7e3226e4ed870703ebd8a67d1b66ad3920e14b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-07 13:16:14.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/795 : SUCCESS in 0h 42m 37s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":6},{"id":"60f04a5f318fd20dfdeb9f81a3afb64ba1b0cfef","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-07 13:16:48.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/438 : SUCCESS in 0h 43m 13s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":6},{"id":"7e41a13c0a4f2ed52413ce15b665d95d4f9b7016","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-07 13:17:06.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/523 : SUCCESS in 0h 43m 32s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":6},{"id":"445c810570647f0f19f81742ffcb2ce6303aed3c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-07 13:32:48.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":6},{"id":"1e5cb97809334830ea29e70c07cbbdcd7f29cf8d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-07 13:33:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/48aa4f4793064d158ab62ffde1f94fa6 : SUCCESS in 58m 16s\n- cinder-storpool-tempest-iscsi-multipath build/9a7123eec2ee4075bea771b98d13b187 : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":6},{"id":"a39628d2e91288083577ce873437ca579537eaa4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-07 13:44:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4715/.","accounts_in_message":[],"_revision_number":6},{"id":"d6d7f41856ed47b1fbb75d3fff9624479b31ac4c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-07 14:05:55.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8288e51d4c1543f78a1ffef97ab1f313\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c40bc4d24354bbca427f3990798741d : FAILURE in 51m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/28368a337e2d4d84add85d857aa7a406 : FAILURE in 1h 32m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1c803b90e52e45b8a4743237834781f9 : SUCCESS in 35m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b5e3885bbd0e4588b669eaf996bb0a83 : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":6},{"id":"07af75e91dc5580accc3fa3f46a7800f7aa974bf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-07 15:07:04.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-6/FC/ : SUCCESS in 2h 33m 27s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-6/iSCSI/ : SUCCESS in 1h 46m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"f11ad98c5f17c7ebbee2f4dce320ddc8ea09ab77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:48:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9672b501335c4826b8b65a7e2598fcd2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c14626107444edbcbd43c9a5e8064b : SUCCESS in 7m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddacd6b4789046ceaefd29127266adcd : SUCCESS in 12m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f9424c6fe854b66b46fe2ce096b19e0 : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19499dcfdd3e4b68ad2d7104828a040b : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ccb047e5ccd4d3b87992349b2d7f247 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/62400027cdbd405b9e35dcbb263dacc5 : SUCCESS in 1h 11m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ea241c23f014b45839ed68058a212ac : SUCCESS in 1h 08m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/913deb7bc4504db694b1baaa5a5e9979 : SUCCESS in 1h 35m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d5bdf586b804296a807a26d933896eb : SUCCESS in 1h 08m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae28458f04274475bbf1577390fae870 : SUCCESS in 9m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bcd33ea8caed4ec7a839755b021216f2 : FAILURE in 23m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/357fec14bf0f4430995c3060fbd5dc1a : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c05ea75be45e447d86cf07b60e2af79d : SUCCESS in 6m 44s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f0dc61e125f54fa59e2e800d7514ecba : FAILURE in 16m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ed9ba0e8d48541d1b7c24137584ee980 : SUCCESS in 7m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2e3ca4a863e04137b2e404446bb5f113 : SUCCESS in 6m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/36e4df1f255f45f08f24f887781d4189 : SUCCESS in 8m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/11b4c924e2924c539b09a33057179d4c : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/90f874bf082b4dc8ac9a641fd100796f : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83a76099c8b542bd800e84f86febef7a : SUCCESS in 1h 54m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c57fbb5e1ed84d64a5037cbafeafe884 : FAILURE in 2h 19m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c6427ea38ae04ed09251feb9746ed3a4 : SUCCESS in 1h 42m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9249af93a4fc4c4b97aebeb5f60e8867 : POST_FAILURE in 3h 09m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d6e124752aa94a3290e495563d139404 : SUCCESS in 2h 05m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/31697d816b8d4c40ab17560b46df2692 : SUCCESS in 37m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/659cad9198404bbabf75d6dc85ee641d : SUCCESS in 1h 19m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e4147aca91c4a90af5b1c7902bdbec7 : SUCCESS in 1h 50m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/77a1279ca853492280939531017e24d5 : SUCCESS in 45m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8a22c5d893404499a67f47787c2a2bb8 : SUCCESS in 2h 03m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b273509fbe4490195e23e914f229e1a : SUCCESS in 1h 10m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"19efdd77f710a03c1b8d6c665d90f147a3928559","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-07 15:59:15.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_07_2026_15_51_40\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"5d5a487c5604529f2f25cf6938697a6d673db5fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 16:16:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71098a33248c4b8ea82879fd3622c024\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00beb98740ce43939db272fcfcb97214 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4d88772c6bf94a59bc9b9873caa29c97 : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ae96a71621ec43e1876ca403fca8a31b : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c56998cb015a6de4eb59453a24b7879967f7246","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-08 13:03:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c0762f8e3a70fcd9f183322d859c3f5a674958ae","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 19:26:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4298/.","accounts_in_message":[],"_revision_number":4},{"id":"51e694a8a25bbce9b043106adb22b0a14d28cb83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-13 14:14:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a061a4532e8e3bf4de57f62e5fc3d230a8548db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 14:40:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a24d32839c64af7af9b75c185874a55\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5428a5dffd946dfbe44734b01ed44e1 : FAILURE in 16m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/422f2ade4dce4dce9115feb5aec79e39 : FAILURE in 16m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb9e50d7559546f7954356d8ec82438b : FAILURE in 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2de6cc5cb35723ae7bb091b6207d5a189e0e25fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-13 14:41:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"65b89e28da2b328d22e0624a9d86caed212281e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 15:08:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3a59a279c14108ace99fdc1f6999be\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1cc5768e2d9940adb52fbdb95da19d5a : FAILURE in 18m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c234e3bd4c244e8bbd1daf433829a9df : FAILURE in 17m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3ef42e908e4d40e4af74ba2f5e6a3cb2 : FAILURE in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f0319a2f7433f14199cbb0c9dcd39626f84b759d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 15:09:57.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":7},{"id":"ee5f1d7f00f1a1d01a9e61afbb4589e4a5230026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-13 15:10:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f7a286a04f472b6c7e19b60729f9ae25b7e2ff63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-13 15:11:32.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c370aed0a8471708af405a4a816765fc8d54c3e0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 15:23:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4783/.","accounts_in_message":[],"_revision_number":7},{"id":"2e282cf6f7e7ffc4709be32adad360ddb79e7407","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 15:28:56.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4784/.","accounts_in_message":[],"_revision_number":8},{"id":"509dd891f543bd039e43601527d01d63dcc750ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 15:34:36.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":8},{"id":"d237c3e67a2585aa6c48e7673729a24817c796a7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-13 15:45:28.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/f038f2b034504bfda3f2ce99b4e347f0/","accounts_in_message":[],"_revision_number":10},{"id":"92836654db8dd4e5b50b243d713f8097a1229e05","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-13 15:51:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f038f2b034504bfda3f2ce99b4e347f0\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/46d8b4093f2945f988a02ba3557dafa8 : SUCCESS in 34m 59s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25de5524c6f949f1b8eae233abd00797 : SUCCESS in 37m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/40028c1e49e347c8ba005a426736e90c : SUCCESS in 30m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c0a474b42f514a969c5fca08ad3857ce : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":10},{"id":"5454a621b732d9e7b95d0bfffbf0c389f556c24c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-13 15:58:44.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/864 : SUCCESS in 0h 44m 0s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":10},{"id":"db6a8216f6e4ac5e541d4b2f6080013a8d817026","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-13 15:58:52.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/508 : SUCCESS in 0h 44m 20s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":10},{"id":"7c6ed5be98e753357b6c21b470add695243e39bd","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-13 15:58:58.000000000","message":"Patch Set 10:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/595 : SUCCESS in 0h 44m 35s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":10},{"id":"aac88b23bacf3ef059cdcc57f29c90c5453f93bd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-13 16:03:36.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/94d7db3f05de43029b27757a316cf639 : SUCCESS in 48m 20s\n- cinder-storpool-tempest-iscsi-multipath build/bcc322685b0e499b84e74e2971e78dba : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":10},{"id":"e331787c1bc9bfea8a28f46a92874da7145d1601","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 16:07:31.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":9},{"id":"50d29404b44b098993398c363724d182fb90267f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 16:08:26.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":10},{"id":"ca7f828f41881eb54dc07697fca74e0a292b3ed9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 16:28:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4785/.","accounts_in_message":[],"_revision_number":9},{"id":"dcdeca44fc216d3aa9ab97ef702997311a7f1765","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 16:33:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c8ea02f53964eea8fdf4f31919609b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45c070d6c0e64d5c883575edc99c7b10 : FAILURE in 19m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35a75c520776458ca60b570e756292d8 : FAILURE in 16m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ac6a64dcc23f4c378e303cdcdc0932e1 : FAILURE in 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"426cce2795d38a22a8e6894c0c13c4cf39ef6bff","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 17:23:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4786/.","accounts_in_message":[],"_revision_number":10},{"id":"39752d13f575f94bb82208f2741cfc498d021d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 17:51:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a467644435da4366af741eb804ac4866\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e19d077c6c0486794b98b95339ee18c : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a47588f72d0f4430a8da17c86dfd8ec6 : FAILURE in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/260d46e6d3314e938841bbd46d05b034 : FAILURE in 9m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e0485b7f88c49e5b3f09ba91947bd52 : FAILURE in 14m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e3c7b6531c436bbf7ea1b01f0e60e4 : SUCCESS in 14m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/24189acb72a74ceb9a708abb68cc5213 : SUCCESS in 39m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e874127aa40a4e5c94f66a176593519b : SUCCESS in 1h 11m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96c845b8ad884c9d976a8ffc8a155097 : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74de98d1e6ca43d5b368c018e95b578c : SUCCESS in 58m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7b3e63659f24f76af4c74900588ff0b : SUCCESS in 5m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/490286e45bc14a9d96b1b65cd97e5e92 : FAILURE in 16m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/773345986e3249088d98ca388cbaa434 : SUCCESS in 3m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1e37c1d93914d6c890ba8a2202c45f3 : SUCCESS in 3m 23s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a4db0922630140208f97ee0dba4755a5 : FAILURE in 14m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a9937edf75a4c77b830505635b1d3ff : SUCCESS in 7m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8c4c2264340a4541baeed0d392524abb : SUCCESS in 7m 07s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/b3011ba03e444ea78a4093c95686f2f0 : SUCCESS in 6m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9b7ff1642caa460ab107710131b9dcc8 : SUCCESS in 37m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4f0d01e55f24c73b504bd4a61ded4a0 : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf5002d695ca4839be055be70c6e333c : SUCCESS in 1h 28m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e4e7945677e54571a49951655549e791 : FAILURE in 2h 05m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a068858e982f469da33d60347df4d45c : SUCCESS in 1h 40m 16s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/721625d5980842e6af61b28ce2fd0a30 : FAILURE in 2h 34m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/496a2ce750794c038990b7cb849940bf : SUCCESS in 2h 01m 10s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/cfb3f0d5a6424cac9e4c6d64a9ba5613 : SUCCESS in 33m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd7d7e7bebd241f0bb761b1339476ebe : SUCCESS in 1h 25m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b80dfae295a4093bdb24ff1eb5eea5e : SUCCESS in 1h 31m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b79cfe0aff424725aad3a34388ac5889 : SUCCESS in 41m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e531b6e71e6842558db9a0da52fa71ab : SUCCESS in 1h 27m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e579e65ea5a4446b01f1b78ac86b8de : SUCCESS in 1h 17m 26s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"98d5c067fde7a7d0b9b3295fbb858e5924572ce5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-13 20:43:44.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-7/FC/ : FAILURE in 6h 28m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-7/iSCSI/ : SUCCESS in 2h 19m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"d6f8da68030a2ea608f588a7fbad3faea5e85200","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-14 07:17:07.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-10/FC/ : FAILURE in 16h 5m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-10/iSCSI/ : SUCCESS in 2h 28m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"60233031ec45182741bb473081258b13b05d043f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-14 08:17:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-7-fc : SUCCESS in 1h 10m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-7-iscsi : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":7},{"id":"10f801d0d14acb6b45b18b6a12702242d7392784","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-14 08:42:19.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-8/FC/ : FAILURE in 18h 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-8/iSCSI/ : SUCCESS in 1h 37m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"eaa0814a9288170503a2c95264fab8968db26efc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-14 11:16:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4338/.","accounts_in_message":[],"_revision_number":5},{"id":"3b538e2f8425a2f956c33ec57f5c4228e653095a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-14 12:43:02.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"a91ad44738dbc257a4d04e29f16db3f045b685f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-07-14 12:43:10.000000000","message":"Patch Set 11:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/941/","accounts_in_message":[],"_revision_number":11},{"id":"a3a34b5dc40d786a2d2d1af61a764d51a3d7df52","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-14 12:45:17.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/869 : FAILURE in 0h 1m 57s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":11},{"id":"902ffd41156cb7b8a18e37fc3f9af2667d4b8f41","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-14 12:45:18.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/600 : FAILURE in 0h 1m 54s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":11},{"id":"0cf9e5f93432c3024d21071d0933494ecde28515","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-14 12:47:31.000000000","message":"Patch Set 11:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/513 : FAILURE in 0h 4m 10s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":11},{"id":"51a94bb93ff1d57758e9f9c344f0f17d158d8330","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-07-14 12:54:37.000000000","message":"Patch Set 11:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/993363/11/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/993363/11/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":11},{"id":"bb74180d5c28614dfa585b57c2c334c67f50dff4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-14 13:21:33.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/52feed2bf22e41809a5f7696d513c839/","accounts_in_message":[],"_revision_number":11},{"id":"c3661e5d4f852d1021d36ef47a3feda63e633fee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-14 13:28:49.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/52feed2bf22e41809a5f7696d513c839\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/af69c590816a443d818c034267a37d88 : SUCCESS in 45m 14s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00d315e17cd542f3b0e023777b574da7 : SUCCESS in 38m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/106af5969ddd466cb083b8e22bff1bbe : FAILURE in 38m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/720d270e60304bc19e32089758a88ba0 : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":11},{"id":"b85bfd3949e1550f31628450b20741d047deca5f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-14 13:38:51.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":11},{"id":"4683fbec4bf12ed7b1e1069f39e22bbe1b993e8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 14:05:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/078af1d981bd4d3699ad4abc607099b6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41839860b4aa455c832646b09fa3008b : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cde61ee3a8de4402a02de3f4c3974b2b : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/faf75e4c67b04d2ba0c0a3a51c9bbe56 : SUCCESS in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"44fefcea7d9c5f4b14aaf58c8adc1bceda9149d0","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-14 14:07:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-6-fc : SUCCESS in 1h 7m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"21adde6f334b53a9c0b2c509911591d289372067","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-14 15:12:39.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-11/FC/ : SUCCESS in 2h 29m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-11/iSCSI/ : SUCCESS in 1h 41m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"dcb1117aa4217fd815ff30490e384141463cc20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:30:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98ff5599ffe4589ae4a678dfc1dce4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc685bfc1f2146d29603d6a1893d6fab : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10e50d1071aa4d66bca46ae76378e96d : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb6b962873a046c384f6afed22528aeb : SUCCESS in 11m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd063a6883bd47e6a75df2395a0335da : TIMED_OUT in 41m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a3cad098224fda82ec45faad85d56c : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/7ef47a3ce59f4b37a6275101a7b368e4 : SUCCESS in 1h 10m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f03bd2bb591245298c45beeb65a31de9 : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa967f9881d4441c888a3e12079510de : SUCCESS in 1h 27m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b249792b2db34111be4bbd470485eda5 : SUCCESS in 1h 03m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e08ac07173854df9b67fce142db834b9 : SUCCESS in 9m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9f657878c79f4bc29e2bf0e03cf261e5 : SUCCESS in 17m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3759d24a1a6407496d31c0217365557 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d90bad071b1744d1875483b437aafa60 : SUCCESS in 6m 51s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/35ab27899b564bf68dad0da97490b1f0 : FAILURE in 22m 32s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/50d119ad818f447a807bbf1f5de4f9cb : SUCCESS in 5m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a38b02fe65bd4ceabad6060dab9743bd : SUCCESS in 6m 26s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e9a2f3a4939d4bb38a4ef664bf9603b1 : SUCCESS in 8m 42s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aef490f4aac048709bcf3dbac25ccf35 : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ca632244ce94f75a6a9ffb60cb45361 : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a475e86ce3e44d6bc75ddeaad70a419 : SUCCESS in 1h 27m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/30c4f2f6deb44772b8953059d33922fc : FAILURE in 2h 04m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d1142b18e06245998c8c3e103e118fdb : SUCCESS in 1h 41m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a4743e43d06b408d87a188dbe36bd574 : FAILURE in 2h 41m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e20a06e335664567b2d0009e705d3719 : SUCCESS in 1h 50m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/074bc105af8a461f921114d6ba2b0fa5 : SUCCESS in 30m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7c5d85d2e219445a81966f323fb51e67 : SUCCESS in 1h 14m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e0ea2f94f6b464fa757725bba5aa0ef : SUCCESS in 1h 02m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38e736ddb94d4cf38248783617c65f17 : SUCCESS in 27m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eaff8ae8eb994c7d944397b118ead32d : SUCCESS in 1h 41m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c0ce823c1ab467c91a8387188cd62ef : SUCCESS in 56m 20s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"52b9484eb527ae662beef0abc670155b7b62b242","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-14 15:40:37.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_07_2026_15_33_52\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"e3507e5bf6626d0789041e1802b946ce50de4e01","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-14 15:47:13.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4792/.","accounts_in_message":[],"_revision_number":11},{"id":"9ba9f61d233ddfc19e598ca6cfb3377237663b3a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-14 16:08:01.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-10-fc : SUCCESS in 1h 4m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-10-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"1b5698626bcd738a4f35b5342cfdc5c93efa3843","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-14 21:16:54.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-8-fc : SUCCESS in 1h 15m 50s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-8-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d6af56d09170311f2da9860352aeb57bdfca6077","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-15 05:59:54.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/609 : FAILURE in 0h 2m 38s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":11},{"id":"34882c189e0fd7226cd698e759b8fc775af07b75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-15 07:53:24.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/181f1b503bad40939feedbc6e1746013 : NODE_FAILURE Node(set) request 200-0000001922 failed in 0s\n- cinder-storpool-tempest-iscsi-multipath build/16ae9141dae04570b5ec6d20d85a575e : NODE_FAILURE Node(set) request 200-0000001923 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"3022e8f442213ea412466eec4535717b3c1cad63","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-15 07:59:43.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-11-fc : SUCCESS in 3h 14m 51s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-11-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"501fc2de31e86ea56ecc3ce580cdbd1c4e3e7ccb","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 13:49:18.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ffebc37c8d992aa544a449afcad1009ae44794cd","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-15 19:52:45.000000000","message":"Patch Set 11: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ce65eccfd9ea3d25b624c53d2a67c675cd93ddbb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 03:13:15.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4607/.","accounts_in_message":[],"_revision_number":7},{"id":"77fd36b502ea9ce287dbc2e3c6392d479b83853e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 03:23:01.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4608/.","accounts_in_message":[],"_revision_number":8},{"id":"77d236c353998963a832ab407ae7d493625f703a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 03:31:45.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4609/.","accounts_in_message":[],"_revision_number":9},{"id":"90f26878976b6e92656471ebc59b9dd7ad483e59","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 03:40:07.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4610/.","accounts_in_message":[],"_revision_number":10},{"id":"17c0a22ee05cdb4df9beb6520f294473a9a6c2a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 04:54:20.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4616/.","accounts_in_message":[],"_revision_number":11},{"id":"b8d8f9e1b3a2b141efba6195a8d9c4c8e1b19452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-21 10:43:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"ee8fcbfbbdd6cb6ae3e14ac5fe74dacf3d44f552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-21 10:46:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7d025bf97ae482b9c183bee5d302f3d7f93aab99","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 10:55:01.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4863/.","accounts_in_message":[],"_revision_number":12},{"id":"bff9914fd60362a8e370d86ddbf4d5947f9cdb5f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 10:59:59.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4687/.","accounts_in_message":[],"_revision_number":12},{"id":"7327eb8ee516113bcb50ef6f29d0b7894bad781c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 11:06:32.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4864/.","accounts_in_message":[],"_revision_number":13},{"id":"529f64df1dc5bfdb65709748692cd6d6c64dcf37","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-21 11:09:55.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4688/.","accounts_in_message":[],"_revision_number":13},{"id":"c08dcb903b9086cedb373844e94b59e9bf6a2352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 11:33:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650fd8fd931f4d72a3c85bab3c219701\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eecabc6d15344eb390874c418646fb11 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30b6c766e71b4613a11bd86b0f1f863c : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/630256928bf84746bb131265eb73524e : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5419e7a60e7c353c41599387ff398056aca6b2bc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-21 11:34:41.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/993363 : FAILURE in 30m 24s","accounts_in_message":[],"_revision_number":13},{"id":"05ae9c06145478392ad182f7a83400995c1c12cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-21 11:40:11.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":12},{"id":"d4779f13db2f824c87a93a3d975dbaed14fbe652","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-21 11:43:18.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":13},{"id":"bddbe4c2acae325dff18c7576d03133cef3eab07","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-21 12:06:06.000000000","message":"Patch Set 13:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/570 : SUCCESS in 0h 45m 58s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":13},{"id":"74f8c359b55ad9968e03d3015ef1d1c897d44830","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-21 12:08:03.000000000","message":"Patch Set 13:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/925 : SUCCESS in 0h 46m 4s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":13},{"id":"5d20eef113ca0b06140ec7b614e4285c42583d99","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-21 12:08:12.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest build/7f7ce7849f434aa686755cf7e898fea0 : SUCCESS in 46m 27s\n- cinder-storpool-tempest-iscsi-multipath build/164d6d3e29ca4b458e6b7e81b02c8260 : FAILURE in 50m 19s","accounts_in_message":[],"_revision_number":13},{"id":"d1e295c34553fe44f923fb6d17c13bb48d4448f6","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-21 12:08:31.000000000","message":"Patch Set 13:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/662 : SUCCESS in 0h 47m 40s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":13},{"id":"cdb8a335dc3bf5b85ffe9ddf861b1901d6af882a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 13:23:27.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba310a54b04415ba7ceb981991286b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf067183558e4e20b8c7b6f3144697dc : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c442705e49349e7b37afaa5e98c475e : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff8977a7d690421d907cf9c57ef8cfd5 : SUCCESS in 10m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c072457c190a437f847e2ba38329d7e0 : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e04c467cab45c18310cadc414124cb : SUCCESS in 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/4643d89f0d2d4c26a80b3fecf7fbd704 : SUCCESS in 1h 08m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69e2ce5364b8430184ad723ded3fa76f : SUCCESS in 1h 19m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d02b1a2ab8564d2abe2e28f3eeb92b4e : SUCCESS in 1h 17m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ecce855bda4493f87001725ec0362bf : SUCCESS in 49m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d437778f13b347b686b071c5aba30e07 : SUCCESS in 8m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6f2746dca9714b3a8b4bbe8fae887de1 : SUCCESS in 28m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/408688931e014e68b587a6a7127bab3c : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8cc48b6a23844bdb9ab1b187c3221bdd : SUCCESS in 6m 24s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/e6eff10d111943f8aa583becfb7be758 : FAILURE in 16m 02s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a79c0f3f42d74540954a5f2f98d895f1 : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/833f029ad3ec4519a88ea7a323e6e1f9 : SUCCESS in 6m 08s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a9f4ba067e524d96ad5c096328a35e24 : SUCCESS in 8m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/26926395526449698427c9c2da2f989f : FAILURE in 55m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/79c3d60ac0f04e7cbe565bbac3f72db8 : SUCCESS in 6m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a98e619d118d48caa2528b1e203002f6 : SUCCESS in 2h 05m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa55857f70604797aa3c86ed8d8a76f0 : FAILURE in 1h 46m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1acc78a2fce24935bff4eeebad3aa47a : SUCCESS in 2h 08m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/a23fb4e222f143dda86f3679ed296313 : FAILURE in 2h 35m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4c4a6988b78b414b95d5080f9d0de382 : SUCCESS in 1h 35m 04s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/59e7e315f7ab423c9a9b99207a924dc5 : SUCCESS in 34m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/11b18fcf6e524046824e6f298d3af1fc : SUCCESS in 1h 13m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/965f9dbc84f0458fb9b28367d0b0776d : SUCCESS in 1h 15m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32013f32a7eb4c8fbf4d614fe03d7d65 : SUCCESS in 40m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88625d90f1e040c3a95ba4aa3039652e : SUCCESS in 1h 40m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2daf71317dba46ddb12b524f0ef3e8f5 : SUCCESS in 1h 01m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"b8a48bf65c2ce2166d5e3ca6be01f388c141ca03","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-21 13:34:14.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/55d0e201726945f3a58defd663ab2dcd/","accounts_in_message":[],"_revision_number":13},{"id":"a98709fb1448a4e2c2612188cc9d66dedbe31b2e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-21 13:34:17.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/55d0e201726945f3a58defd663ab2dcd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8fb85b1cb4dd43449e66418d3226e37f : SUCCESS in 44m 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0eec350116d49ff8324ca36ad2bdf05 : SUCCESS in 37m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e09b154f7c1f4dcfa3a1e527c7522df3 : FAILURE in 47m 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2a0185fe49034939923ddf65de81973d : SUCCESS in 40m 15s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/df9a09e199454cc792613773fbcd138c : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f746cab021d9b034b30281cfef393a85469a30ad","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-21 13:40:22.000000000","message":"Patch Set 13:\n\n(20 comments)","accounts_in_message":[],"_revision_number":13},{"id":"10de690fe55b7c6377263fc10f5414b89e576745","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-21 13:40:28.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_07_2026_13_33_29\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"80e3b1685865c8f59d11532b9ad3f94feb092668","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-21 14:02:03.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7ee2b8db7e61beb65a64cbb5a3d0985362795e57","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-21 17:47:47.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-13-fc : FAILURE in 15m 42s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-13-iscsi : SUCCESS in 1h 7m 25s","accounts_in_message":[],"_revision_number":13},{"id":"451f881ab86078502027c4cd06ea4b5ca7104c28","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-22 05:50:46.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-13/FC/ : FAILURE in 19h 3m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-63-993363-13/iSCSI/ : UNSTABLE in 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"bf118f04ddff9420e10856d14c88d4f3d8e67c40","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-27 05:24:46.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"c72d71c6d64f68b53a4a09a65dd4415c7125b1ac","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-27 05:24:53.000000000","message":"Patch Set 13: Code-Review-1","accounts_in_message":[],"_revision_number":13},{"id":"4520a5d36ca3adcfa6556ee5395ba0086c09a6c4","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-27 06:18:03.000000000","message":"Patch Set 13: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":13},{"id":"870f2d2cf4e6e093c95c9fbebd5a3cb72f13b1d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-28 12:52:01.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"7dd71f546d85d2b514ca85d2f6964c6872595f4a","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-28 12:54:23.000000000","message":"Patch Set 14:\n\n(13 comments)","accounts_in_message":[],"_revision_number":14},{"id":"046c429324c14004e520f482a891cbe06c434d04","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-28 13:00:55.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4757/.","accounts_in_message":[],"_revision_number":14},{"id":"e52f9f1b2ac0612265bee1ba8e41794620e9bf9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-28 13:02:52.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4933/.","accounts_in_message":[],"_revision_number":14},{"id":"ec1d3f78cfb91303a335f902ffe2f7e8b2ce5e04","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 13:15:04.000000000","message":"Patch Set 14:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2ed1081b766c444ab1c0511b0dc57e6f/","accounts_in_message":[],"_revision_number":14},{"id":"8bef5d89a3c7aaafa72414830a8fc4710bb14e7f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-28 13:34:27.000000000","message":"Patch Set 14:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2ed1081b766c444ab1c0511b0dc57e6f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7336e404d0946c2a9a750741f16ce57 : SUCCESS in 39m 24s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/01ccefc28e234d8a93ac27716a34fc15 : SUCCESS in 32m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9cfd9d7c81a4441487513fd3fadf4699 : SUCCESS in 41m 54s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff823cfbf42c4d2886749cbe598b1afe : SUCCESS in 32m 33s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c2a78189c634bc28ef0598cff224b0a : FAILURE in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"190f040a660857f658897cfc358ead4a0994c014","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-28 13:41:13.000000000","message":"Patch Set 14:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/728 : FAILURE in 0h 48m 51s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":14},{"id":"640e4fde46375ebbdb563e4add8579f40cc55c82","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-28 13:41:53.000000000","message":"Patch Set 14:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/636 : FAILURE in 0h 49m 31s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":14},{"id":"e278e61a87964abb8ddef7c94498cbe7d963e803","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-28 13:47:44.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":14},{"id":"e02bf22fc89485a5c94b65808f74ecca6e3b82bd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-28 13:58:24.000000000","message":"Patch Set 14:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/994 : FAILURE in 0h 57m 43s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":14},{"id":"7b2f863a9940fface4698c37f108d6f3e9449c7f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-28 14:50:44.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/2f243eaabd2a4eb8baf272bb615d49f0 : SUCCESS in 45m 56s\n- cinder-storpool-tempest-iscsi-multipath build/db5e676d463f41c18fec5ecd1b6c90d2 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":14},{"id":"434231a0bc4d1c5046baa90070b4fabb4acdab8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 14:52:11.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fbcbe87d9454b19a4e76ee2ba2a237d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e70a724247c14553b65ee16a51689ea9 : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aee8ef1d957d46b2a0e9d9c6a77be61f : NODE_FAILURE Node(set) request 311c761843a0426db032c8e5bccf3fb9 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a0f1899300814ae3950b7d9b7ed46083 : TIMED_OUT in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a39b4091a7522ee14773b78625eb21feca49ea93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 16:12:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c74a295bc6c5459bbfb3b1973c04a39c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb38dbb366074cdca4e11568003fd134 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c7fa0e5bef241b7991a173b9ee0400c : SUCCESS in 11m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efce57edfa5841eb85bfab62bcd51713 : SUCCESS in 10m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e49718220e8f49818348ff4f47b07ce3 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fca3ddaadf4cd2a4ab1dc46627c576 : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/989315c0410043f8817d430e6ae43fc0 : SUCCESS in 1h 01m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/60e24e1768014ec397917210162fba02 : SUCCESS in 52m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec56ca006cc2459aad7e3878ec85ff5e : SUCCESS in 2h 02m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf8d62efe6148059e57ccf22117de80 : SUCCESS in 1h 00m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc381ed098d34dfd85c6c3d3930c159b : SUCCESS in 10m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e21d8f6a27fb4a63872029cc253fde97 : FAILURE in 26m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1325dfc8e6cb46419aca9371239df9f8 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84d7b49ed9b544c58074a313c394561d : SUCCESS in 8m 32s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/373e9fdf5bb345afb5c2da6e168e50f4 : FAILURE in 14m 50s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b097562b1172440b9ad0160b82070d31 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/80260b9ca6994fb8ad99908923ddcba4 : SUCCESS in 6m 19s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/91164f6fc21e455989dbd7e4bf1e856a : SUCCESS in 9m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/71203ce11b2f496e89fd1007dc71630e : FAILURE in 54m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7727a0c341e4439a55077ac48e9b9b4 : SUCCESS in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/32a7560206d445ba9888ce9e113dddb7 : SUCCESS in 1h 33m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/75b5c76f58934e309aeb5bf3a9e34859 : FAILURE in 2h 44m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b2218651c1774681a03d30a043ad20fd : SUCCESS in 1h 45m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8a281d12f1c5443282771ba73af74165 : POST_FAILURE in 3h 13m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d74856e1e7d446de8f2009454d8b1c28 : FAILURE in 1h 39m 45s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d2df6126d1bf4a008e9a70436d64c63b : SUCCESS in 37m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ebb716b6d82d47a5ac5b486876ea6ec7 : SUCCESS in 1h 11m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a15f001657584fa88f1b6e9b6d16d181 : SUCCESS in 1h 39m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48caeb6468c44971ba86daf04070d3a3 : SUCCESS in 47m 48s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/14224394dcbe4b93b62083fd5b2c9faa : SUCCESS in 1h 40m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/728d7ccdfe9a414ca274212701e51e35 : SUCCESS in 57m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":14},{"id":"8303fc260efc5a51d24daf70b60b402e687e6285","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-28 16:23:01.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_07_2026_16_16_07\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"c5e6432fee33c42ba7e672c0e1b9ea0769f17c6b","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-30 10:04:46.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"fe2be705362aa6fb1fa5c80784a6bb71a4c8655a","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-30 10:07:36.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"04b2188cafb69eae84991ab8d60a1211d9d98428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-30 11:26:21.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"4f9081091fbe3a3383dc87e36e60167937228a78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-30 11:38:28.000000000","message":"Patch Set 15:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4784/.","accounts_in_message":[],"_revision_number":15},{"id":"337ccdf49f4d0f4e582dd97f4880317c6357889f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-07-30 11:56:31.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7321c4def2e9a7529f77c48d03367b13a52040bd","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-30 12:00:12.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"2a3f0df560f531378daec5bbabdc5d3c6678f869","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-30 12:00:36.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3229b948cfc8705eac16c5012b0f6587153fdc22","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 12:01:26.000000000","message":"Patch Set 15:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a9967dc637284205a7c9631029868bd7/","accounts_in_message":[],"_revision_number":15},{"id":"cbb927b780ae549fdfee771f664da238c228a0ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 12:15:36.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a9967dc637284205a7c9631029868bd7\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d7088a655f8f439ebbc3fa06ee1d935a : SUCCESS in 39m 54s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fd67f72f05d4f448c4119d2fa930815 : SUCCESS in 38m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea17d35bd0884de58bbf46a7490f438f : SUCCESS in 44m 46s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3b6e195c159d4373b10bd6a861917486 : SUCCESS in 34m 34s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/26cbe02fd6814712aeec55ac2d802397 : FAILURE in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"caae188c7df4c6882c76df8496fcae8d5bbdc1f1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-30 12:17:10.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/75fd7fc2d49247b4b80abe7ed8470889 : SUCCESS in 47m 11s\n- cinder-storpool-tempest-iscsi-multipath build/7cba6dc730f64931ac4482dc0330c5db : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":15},{"id":"e4737492c8c161c1ee1f5102844042457f39e33f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 12:20:36.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":15},{"id":"69c343e2ab2df43f906f636c50b7aa041fcf54bd","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-30 13:01:02.000000000","message":"Patch Set 15:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/663 : FAILURE in 1h 34m 21s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":15},{"id":"2b864a47a4c31e361e6bc295d8642c06e36ebe4a","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-30 13:21:13.000000000","message":"Patch Set 15:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/755 : FAILURE in 1h 54m 29s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":15},{"id":"41de6b2cc24dd491b3b2fcaa788723188bb71089","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-30 13:27:35.000000000","message":"Patch Set 15:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1024 : FAILURE in 2h 0m 57s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":15},{"id":"30b89ecdbdc6eabd0857ef7a2e4b57ce420f6312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:01:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c20a089f1ee464595c11ae8f2928b02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381f16a16f4044cfb5cbce8932f91cbe : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c475f819e3346de95961122e89fc1d1 : SUCCESS in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c851860f6546440d9be29babd739bdce : SUCCESS in 10m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1d51936fba348dbac512aa7f78bcf85 : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b833afca06413890f766173ae420f4 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/6595967147884a1eaff9f8ea1ecbea3d : SUCCESS in 34m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c99cb23e84d4dddae3834aff730b2e3 : SUCCESS in 1h 23m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8461106baa546d780fca1acad726899 : SUCCESS in 1h 05m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe132481e36f4db9bb4ed8e6aee34a5c : SUCCESS in 32m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e74de9d79a54acb9d20b1cd82dc9e89 : SUCCESS in 5m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59ff014cba4b4821bd81634184a1a831 : SUCCESS in 20m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c6a9457122414e8083b71301fd3064b9 : SUCCESS in 3m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c8c27bbcdd8404e8f4bfd06f1e9816b : SUCCESS in 5m 48s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3ee6809581c943148c816a8edb06330a : FAILURE in 21m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5fbf291513d8407a8a99c7a5c68d1e94 : SUCCESS in 5m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/af8ddd00f675416f93c19bf6562d5936 : SUCCESS in 5m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8d5c8871e2fd44669ec9dc39d639c67a : SUCCESS in 8m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3561f3fa9866436fac713bf49959de9f : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dbec70e8d40d40a3a89dc323bfedac1c : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e8e032416d344e0bd0e2d58c6d604a0 : SUCCESS in 1h 32m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fcfead76cf724f5895a0f859c798c580 : FAILURE in 1h 50m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3b037f36b6b74439897aaa80343bac0f : SUCCESS in 1h 27m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8c7f82da200c4ed4a8bd015b8efd1b7c : FAILURE in 2h 29m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/13189835357e43ffb26a7f4457cd0fae : FAILURE in 1h 31m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5c420d3208e647a4b60c069450f81a40 : SUCCESS in 31m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/de90a8b3ccd1402f9023aebe8cca839f : SUCCESS in 1h 21m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bee95cc60c3f45d0937ee4257388f329 : SUCCESS in 1h 41m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aaa9f7ad65b04fc79583529b6f2fa7ac : SUCCESS in 55m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cce89a9310c40edae6200e379d217da : SUCCESS in 1h 28m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ead0568cb4e34499a24adca97f5b2614 : SUCCESS in 40m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"83f9dd574e4c2594feac9983f34436cbca69c07d","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-07-30 14:02:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4263bdd0498f1e7e042706f688844465e454139a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 14:12:08.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2026_14_05_03\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"a2cb67f2cb9f36e1412cf90393c68a962c3a0dcd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-30 14:32:49.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/15/upstream-check/cinder-tempest-plugin-ontap-fcp/be368cf/ : FAILURE in 27m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/15/upstream-check/cinder-tempest-plugin-ontap-iscsi/c8dd1af/ : SUCCESS in 28m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/15/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/08db522/ : FAILURE in 28m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"745fc6a8d7ee88c05fc0f124d47bc1f4680773c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 15:46:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3cf401a8aba4528b86b87f8e6cd3f8f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a8b9cdf6c0a947209019255c339c0a84 : SUCCESS in 26m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b34ef6f5f6fc4c1aba3154fa6766a070 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/50bc897cba9849c6a10440ddd4f10441 : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a576a446cf74ce9aee2ef15cc920df51b741645a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-02 04:50:00.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"3e88ac7e96c1daf68fd20058abea348d6b1ee20e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-06 22:18:51.000000000","message":"Patch Set 15: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"09685c9fc45bd54eb101d8908b5ccdb095fb1e33","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-11 13:36:55.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"75de8dd113635bd1075e017b5541cb6f9e74d3e7","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-12 12:10:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d40ec45fde69ddb814c85d2ac9ab07dbef744d35","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 16:17:23.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bb16fe90530c6ef0d3eaf14fda127c96c6ec15c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 09:44:44.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"a96c7cb5770f93dc8a493dd404dcd8309c1c954d","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 09:49:30.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"5027d587bbd1d41f30938d94f0cb145227a57662","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 10:02:06.000000000","message":"Patch Set 16:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e9c988777b2d4d9aa8b1605b2ad0a739/","accounts_in_message":[],"_revision_number":16},{"id":"87256cfffb4068b3f2d1fec8f5d94289cb69b9e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 10:12:15.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"75b2918a068d948d576fd826efccd6cfa8fa2e07","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 10:12:53.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"cbbb29f498699ae8d90dc884881446289623bcc4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 10:30:01.000000000","message":"Patch Set 17:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/eb092135a7d545c884de319462d08304/","accounts_in_message":[],"_revision_number":17},{"id":"2c2f600753e0ba461a8d7da972c7caa997990b0f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 10:41:28.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":16},{"id":"79adf56b1585b0608def16a34ee9efe83e729ace","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 10:54:00.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e499fe7ff747b982aa6de23d4b9438\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cad88393a7cf4931b33522054e939c35 : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/309f6eab73f843768268483b01de9e82 : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bbeb2603811047679c5ed2b29006887a : SUCCESS in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3e7c4214c06dbe9e11e06f3ec49bc18349ca9c81","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 11:00:38.000000000","message":"Patch Set 17:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/eb092135a7d545c884de319462d08304\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a78c7e1245a4fa5a3033a2472a8e9ae : SUCCESS in 44m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6ab03f6c948f47db9997da86e56b1301 : SUCCESS in 35m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d787e16f29e24be5bec933e84ae55de7 : SUCCESS in 47m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/790d61697c7248b09baa5ad050aa5e74 : SUCCESS in 33m 06s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4923bbdf3ffc48f6a44bb5a0d1e36ba8 : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8637dd2fb7a5b262415fd28244d3be2411352538","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-17 11:06:02.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- cinder-storpool-tempest build/963a546cd9f546c989be5ff2807f1b2d : SUCCESS in 44m 06s\n- cinder-storpool-tempest-iscsi-multipath build/08022eacb12e47b7a7f687cf232347ab : FAILURE in 51m 20s","accounts_in_message":[],"_revision_number":17},{"id":"1c3179090b5ecf81291969be5b7cbda4a5fefecd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 11:08:55.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":17},{"id":"32eab1c9bf7057abe34968fe849859fa3a358ee5","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-17 11:15:33.000000000","message":"Patch Set 17:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1205 : SUCCESS in 0h 59m 53s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":17},{"id":"b2260db59d3605fb1cf7e775b510d63c3386f371","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 11:22:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"731f13fbecd1766dcb9834727ff59ff9481523b3","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-17 11:22:09.000000000","message":"Patch Set 17:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/938 : SUCCESS in 1h 0m 36s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":17},{"id":"0228ae78482d85f82bc4e020a995ab5aedcd46cf","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-17 11:26:16.000000000","message":"Patch Set 17:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/844 : SUCCESS in 1h 3m 36s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":17},{"id":"cd177bee382eed2e7c15e4153e837d1a6854058f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 12:29:46.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78b5b19994b04fd0935827ea22d710c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25873a63838747dba090d4b9e713ce84 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/447b8a7d65e44ae9844572d3dc56b698 : SUCCESS in 8m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/757bcaf45ea847b08dbaf38b8a0c3c93 : SUCCESS in 10m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/531a386dbffb437bafe1163638fd11da : SUCCESS in 18m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68193266ad2c402b9eca5bfa38b9f697 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/2808efa6e7d04ac183e1b98a3d876924 : SUCCESS in 35m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59707beea8934a108bd0db39004885a7 : SUCCESS in 1h 07m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a51ec7f277214aa2aa8737e638cf4f9b : SUCCESS in 1h 33m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/017fdf96a1874f51943875ba0a9d0a5d : SUCCESS in 57m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db8aec0d3b15443196984829a97d65cf : SUCCESS in 6m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b085ab1a6479493a86c7111a356d55f6 : SUCCESS in 23m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5d06314877fe4e6faf05e07778d9300d : SUCCESS in 3m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1e793c9501e48e09d83b05b6917d748 : SUCCESS in 6m 07s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/3da869df00b84442a1ee9c0eaecc571b : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f2901fcd7d63419c9ce934f385a47950 : SUCCESS in 5m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d51cefcd3c81404bbf5e22fcb5e3fc78 : SUCCESS in 7m 35s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/eaa97a49714d4ebd9b0764f1406f6d5f : SUCCESS in 7m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6a4660d1484d4ebd81946a16961f3861 : SUCCESS in 36m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/534ee2c5a85b4766887507108e0ce409 : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/de3a972e246443df8da90f273897d06a : SUCCESS in 1h 09m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b9d6af701fdb466e9cf812211c04ec4f : FAILURE in 2h 11m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0876b7a1fd6b48b9a8a22fbb10a1148f : SUCCESS in 1h 27m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/94639a28388b472cb49318fc8980e35f : FAILURE in 2h 11m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/607146b2d55049118ef497eb7706968c : SUCCESS in 1h 55m 15s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3d310d30cd2b47019ad0138ecf3b4e16 : SUCCESS in 27m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02ce06f65f134830bb887200f3a029e3 : SUCCESS in 42m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a0f078dabfb040cfbfeaa8c8286208f7 : SUCCESS in 1h 37m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4785fb4c71f4ba3923656804ea7394a : SUCCESS in 44m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cb607f32b41c40758ff8b1ea51525c28 : SUCCESS in 41m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82bab964933a4b6595527468a979e610 : SUCCESS in 50m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"36873c6522586d72a60ccf8028bed1e8c7dac1a3","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-17 12:37:09.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a67f956252b632d7fd4cc10c18f96219a21bfb20","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-17 12:37:36.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"bf12d0f955f12a92bc861cb766d5edd33ef60b5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-17 12:41:19.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2026_12_34_13\nLeave a comment with","accounts_in_message":[],"_revision_number":17},{"id":"7951c5bc6130da2f1b77f4105d13479b0b778090","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-17 13:06:54.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/17/upstream-check/cinder-tempest-plugin-ontap-fcp/4578778/ : FAILURE in 25m 12s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/17/upstream-check/cinder-tempest-plugin-ontap-iscsi/ccfda97/ : SUCCESS in 27m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/17/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/30b64ea/ : SUCCESS in 28m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"0ea48ca81b5fc1d93912b384b105df9d885214d3","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-17 13:44:04.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"30958b64b72d1529c16e00d6df590c65ac5d24d6","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2026-08-17 14:33:27.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e3713d6debf819f160f8a8a8d6615f047a58ac98","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-17 19:54:37.000000000","message":"Patch Set 17:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-17-fc : FAILURE in 9m 24s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-17-iscsi : SUCCESS in 1h 52s","accounts_in_message":[],"_revision_number":17},{"id":"6a78e38f9912169f93643864ebcf269fdce24e92","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-18 00:40:49.000000000","message":"Patch Set 16:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4974/.","accounts_in_message":[],"_revision_number":16},{"id":"0cfd265b73e451c37841c11fdf42ab388cc17c98","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-18 02:28:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4975/.","accounts_in_message":[],"_revision_number":17},{"id":"42883608a1df3e957d7f2034154b60339fbb0dec","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-18 02:53:09.000000000","message":"Patch Set 16:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5150/.","accounts_in_message":[],"_revision_number":16},{"id":"592ac7e0e3804df23777ccda32e1e32a5944c5d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-18 04:47:52.000000000","message":"Patch Set 17:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5151/.","accounts_in_message":[],"_revision_number":17},{"id":"bd1bea34adecd5fcf6c8eff4d31fb9201a14d35b","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-19 14:24:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f530c3b0e72e14bc5b0a43b359e93827b10d5bb6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-21 15:14:34.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dda95c938281d6c5d9f2eddef404c1b2b9fe3c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-21 15:43:38.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"30909c58aaa49dcb3f020beaeb522cbdbeb73829","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 15:58:32.000000000","message":"Patch Set 18:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a1830e3a92f8416281619880544a058a/","accounts_in_message":[],"_revision_number":18},{"id":"77daed6a242e376765c2265661eed3166ec57efb","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-21 16:06:42.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"198ccdc636fdf2aadbcd4f59b66cda6c1e7672a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 16:13:46.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c757c98fb63453582bdb8b8a9acc70b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0c16ec5fd8a34966b6a349191641df42 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/449d5f58e1af4a1faac37971a9194a70 : SUCCESS in 20m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1dedcb10fc43451884f21ab578fa13fb : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a8af3303decc897a11234211ba18a7e58d94d2ea","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-21 16:26:55.000000000","message":"Patch Set 18:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a1830e3a92f8416281619880544a058a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/128c5aa221094476a7ead126d737a6e1 : SUCCESS in 42m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cc9056c000844692ae5a5db6946a573b : SUCCESS in 34m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4126543002354d4093382c57e7cc53d7 : SUCCESS in 40m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ddae52e9da4c4aa7b7143536fde14ea4 : SUCCESS in 35m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/97b2ed8eabab4507a271d2186e3849f5 : FAILURE in 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"113d01abb803f4b209b3ae1570870b4dca0da9cd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-21 16:34:28.000000000","message":"Patch Set 18:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1271 : SUCCESS in 0h 47m 8s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":18},{"id":"ccafb387bcd2e871bf4fc5d8e1c0805829dfd5c5","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-21 16:35:01.000000000","message":"Patch Set 18:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1001 : SUCCESS in 0h 47m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":18},{"id":"43d799ac71f2fb4c6c8e54925f0f54bf32c69671","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-21 16:35:19.000000000","message":"Patch Set 18:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/908 : SUCCESS in 0h 47m 42s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":18},{"id":"828121577d08efcc2fa183f2abffb650a1f0e417","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-21 16:38:33.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/b44066cb522e4299aa4ac64cdaa2616c : SUCCESS in 51m 41s\n- cinder-storpool-tempest-iscsi-multipath build/00451420f0a14653889ac5977c30f091 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":18},{"id":"cc691ecd71b0cc26ebbafbe6805c4113cf7395eb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 16:56:00.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":18},{"id":"023972c1eea1488e8820c0fbeaa4886d550c6cf2","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-21 17:18:52.000000000","message":"Patch Set 18: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":18},{"id":"a7addf6c2272725c6ac50048eeaae202b4c3c1ef","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-21 17:46:20.000000000","message":"Patch Set 18: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":18},{"id":"f6bb7dc184af4cd651f3bd63c95f689ae2a0a6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 18:03:03.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffcfa96f32c142fe808a2139535d2ffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2042ce6a89430897d3626508fbc67f : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72958a7a97a34585b8c8aec4961e17d8 : SUCCESS in 10m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31cce7b92e364281b08f4205f798ab0b : SUCCESS in 8m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/067bd2139840420bb3ee509e4f9b3280 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9431279317e44e798f3c98e9c5eff9f : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/5610fc9214ec44dba2b7506b2ddf0b79 : SUCCESS in 45m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab6464cf50d34d4aa20c38f84bb2badc : SUCCESS in 1h 09m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fef08e8f98c14e7cafa19c363cc1c833 : SUCCESS in 1h 02m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba4f94802ed0477bb614e157afd5e126 : SUCCESS in 37m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f7b60353e2a4396b391c85d3166dd69 : SUCCESS in 6m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d5e129fb9894e6eb196698d948bad77 : SUCCESS in 18m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ce61399d34345d0896d5424f3999f67 : SUCCESS in 4m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31ba120d79314bcfad3eb3cb474f7ca2 : SUCCESS in 4m 51s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/7584ddc86275494aa291af07dbdde01e : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/41b0aaada7d1454686e20925250b9860 : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6459a0b04374427db6c6771fda8b4eb8 : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3381b70ac193405da753d3e717eaeea7 : SUCCESS in 7m 00s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/aae6a5fafe284202b7b033d28e27d9ad : SUCCESS in 35m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a11d5d5c0afe486cb2c1c9bdefc2fa93 : SUCCESS in 7m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4034369433a14e07adba52beb803b068 : SUCCESS in 1h 52m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/252a720ab4834cc59dee2f7a3aeeb1de : SUCCESS in 2h 04m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2b606d440074076bbf6892e59315551 : SUCCESS in 1h 31m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b64eecce4a694ad8b2ce460e7cc746c8 : FAILURE in 1h 58m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/07072660f7124d78908100482a0a128b : SUCCESS in 2h 13m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e0319c766b8a4e66a31b45be8e13b967 : SUCCESS in 38m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5603ea3081e44ab5ae6b617db197af04 : SUCCESS in 44m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1638cd7332274158ad55accdce22b6cc : SUCCESS in 1h 30m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/554a351bd8ec4b08b461101d1dcd9378 : SUCCESS in 45m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/18fb879530394a8b83ca722486bb38da : FAILURE in 1h 33m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85cf9bf274a4da3bdbe1f56180c0111 : SUCCESS in 56m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"e31e74e0999b32e6ba2813e959186213239e452a","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-21 18:20:06.000000000","message":"Patch Set 18: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"4f4c6ceb17e86761fbeded1f4619a1f22ac7cebb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-21 19:02:04.000000000","message":"Patch Set 18:\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"117e5bfd485016f63dd029d2d559dae2f0ffd2bb","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-22 06:25:58.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"ad0fb5b0a854fad42773d1ba6147f7a88cf74f0a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-23 01:45:55.000000000","message":"Patch Set 18:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5215/.","accounts_in_message":[],"_revision_number":18},{"id":"700fe4f056649b36f4407dbfd0ff57812e341ff9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-23 02:23:49.000000000","message":"Patch Set 18:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5039/.","accounts_in_message":[],"_revision_number":18},{"id":"c4c1727ff30e1af01f568de9e083e505b9771f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 15:12:53.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"3570b751baa83bb1092cb0fe1f70489cda12e799","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 15:22:07.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5230/.","accounts_in_message":[],"_revision_number":19},{"id":"114d815a877ad4e86976bf09e55667dd13e3d43d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 15:27:25.000000000","message":"Patch Set 19:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/a667ac1b79b94aa194c5000d1f91db1a/","accounts_in_message":[],"_revision_number":19},{"id":"1f984434312be2e155c2e87c0e63b73d51c4cd15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 15:30:08.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"f23814001152f30deec59037fa091ee30efec20c","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 15:31:00.000000000","message":"Patch Set 20:\n\n(7 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a19d6092efecf1c3caa11422f1fdb84645478017","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 15:39:15.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5231/.","accounts_in_message":[],"_revision_number":20},{"id":"96541159342385a51b15064d71db27b45c1c2c8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 15:44:06.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/896e544e225645a6bf986973f7e5941e/","accounts_in_message":[],"_revision_number":20},{"id":"d683490a2e4159ac5301210417fe7f60216aca89","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-24 15:46:00.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"939bef467d661cb4053bcf6b791c8a621269015c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 16:10:34.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":19},{"id":"93f3ded2b4d5eb19386d2cc4856469b6246941be","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-24 16:10:51.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a36d7f758375deabf0f5226285586dfca5eabb50","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 16:27:00.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":20},{"id":"98ea6268a868d89ea6fb1a4f057b07a1a672a51f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 16:56:33.000000000","message":"Patch Set 20:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/896e544e225645a6bf986973f7e5941e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8f5c17259d0c45129f63d3da85a31628 : SUCCESS in 36m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5dbb9bb8cf78410cbbd7b27384bb6e7c : SUCCESS in 34m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94d88dbee4094849b88790fe63f4a9d3 : SUCCESS in 44m 17s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22c48d487efd479dbd340eafe2acdb69 : SUCCESS in 30m 47s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3f20e06dfc0f435db58b17ea31a1c3ee : FAILURE in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"267f8aa9250808f9178e50542af8bdb0448b0813","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-24 17:08:35.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1012 : SUCCESS in 0h 46m 40s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":20},{"id":"edf7d7c5264cc00899fd11a98fe75ae827392d45","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-24 17:09:40.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1281 : SUCCESS in 0h 47m 20s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":20},{"id":"9f28891e3b577a2cd9ab7c51b6c54a7405929748","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 17:10:18.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"217176995607a15aba68e9e9740ba8ceb61f6041","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-24 17:10:53.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/919 : SUCCESS in 0h 48m 50s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":20},{"id":"a4310af87c4685947835a7575e0d2786f7add2fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-24 17:23:14.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/993fb9e6dc484c2a826bb2181ec147ef : SUCCESS in 47m 32s\n- cinder-storpool-tempest-iscsi-multipath build/63dd998b5b6f450e8b07194f46961fa5 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":20},{"id":"68094ac00aea086110c0ebe68144d5e09d14ab5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 17:45:16.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":21},{"id":"968f363e83600d80112e8a8094b4b2367aa765fe","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-24 17:50:11.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"61c367b09f85980095ccf5e77c2906b089cf06cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-24 17:54:16.000000000","message":"Patch Set 21:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5236/.","accounts_in_message":[],"_revision_number":21},{"id":"4b660fecea18347b863092d843ec7131cad37ac9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 17:59:43.000000000","message":"Patch Set 21:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4ddd70c8cf2f465fbdc3d9695b6e173c/","accounts_in_message":[],"_revision_number":21},{"id":"a937df3795b10427bebdfbdf15ec7e97e6df5882","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-24 18:10:08.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"36d786fe46c833accc296bd2ee7dd428ba4b2a04","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-24 18:21:54.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/21/upstream-check/cinder-tempest-plugin-ontap-fcp/18b4c3f/ : FAILURE in 27m 30s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/21/upstream-check/cinder-tempest-plugin-ontap-iscsi/fb6dd23/ : SUCCESS in 29m 05s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/21/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/41d1b43/ : FAILURE in 28m 24s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":21},{"id":"535a4221d3fc2b6a63d002c9feb047d4f60eac9d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-24 18:31:49.000000000","message":"Patch Set 21:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4ddd70c8cf2f465fbdc3d9695b6e173c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d92d015352d41a181694ad4e70ec5ce : SUCCESS in 42m 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/745c8386c56642afa022cc743f7005b1 : SUCCESS in 34m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9c5c6da7d324c3d9ab9760757d31ff0 : SUCCESS in 45m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/87681a0772b64bc9933fab2d54def2c8 : SUCCESS in 32m 42s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b88696dc6e7143608d7a914aab6b629d : FAILURE in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ab9fd0fc087506ae405fc4bfa0be0d26ee4b4353","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 18:42:09.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":21},{"id":"127c88c7e032ef5a58b8ae2fe711d0935aae1a5a","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-24 19:04:21.000000000","message":"Patch Set 21:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/922 : SUCCESS in 0h 46m 17s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":21},{"id":"73101c4932c140bca777ffae268bdbe429bd7fab","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-24 19:41:34.000000000","message":"Patch Set 21:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1015 : SUCCESS in 0h 47m 32s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":21},{"id":"7fbcc2851ad721cb143e9a89533897f793a2d866","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-24 19:41:54.000000000","message":"Patch Set 21:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1284 : SUCCESS in 0h 47m 25s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":21},{"id":"be0eb994e17a9d09af3895a4eb85b98e0a80532c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-24 19:44:12.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4f5b7009af3d4a119d18545129de1810 : SUCCESS in 45m 54s\n- cinder-storpool-tempest-iscsi-multipath build/100014728b7049bcac0289b249a8ed76 : SUCCESS in 58m 16s","accounts_in_message":[],"_revision_number":21},{"id":"381c7d4841502f7d0884b17b76da5b63895f142a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:12:51.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d8519b7ec94460d872093bae6d52fdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2067d36496432db02d514b54662497 : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00020a46bf2e404f90c485804eda0f9f : SUCCESS in 11m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34bbeafa4bd64dfeb04ff138f9c20d7b : SUCCESS in 10m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b6ae86531d24c4c95714bacc63db779 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2717ed99bab3479db5fb9b9dd9380a6d : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/d4866aa0643c425ab29264e723f367b7 : SUCCESS in 33m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9640446f5935418794c9568bc83b822f : SUCCESS in 30m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/abd277fcba3d498987551ed1a3634631 : SUCCESS in 56m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8060b14593e549628e83a458d67f89e7 : SUCCESS in 44m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcea25a7bc4f49c7832bbd3689fe79de : SUCCESS in 9m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b5ab0e4cf3b8413982b90d6275a9c78a : SUCCESS in 22m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e72636595e5241ca85279ca04b6ad1b4 : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5fca258bf66f402d890d4b50c755e5d9 : SUCCESS in 7m 39s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/a8ef912d086b4d1face1e0f6f946568a : FAILURE in 11m 34s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/25e1dbb920544a3281d382fbf3485075 : FAILURE in 7m 03s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e9e3399828274c0d935108355af1c8a8 : SUCCESS in 5m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bf9fc9859d134bd0b151959444133ab0 : SUCCESS in 5m 18s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/baccbfdb72aa46ceaa0b586777e01dcc : FAILURE in 7m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9e417896f5064ffe95fb79afbda21e64 : SUCCESS in 10m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a77b11d9735049b4a59b64998c92ebc2 : SUCCESS in 39m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9c0542c7af1455aa30c3e0e9d1aa7da : SUCCESS in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ae4513916eba4c57827e4133de53c6e0 : SUCCESS in 1h 34m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d50635f702a24049b05d63a204713696 : FAILURE in 1h 05m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89c0c47c3d46486ea379b9fe3a6d943c : SUCCESS in 1h 50m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/8fdefd3da3bc4728a0f564ff0381ecf5 : FAILURE in 2h 19m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e623dfd8ff294b17a3a745c25bd84b46 : SUCCESS in 1h 16m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7f22d1f40e9f43049a962f8591d66202 : SUCCESS in 24m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a1a4d22d58ac4744b44a6fb916ed65db : SUCCESS in 1h 16m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/50543ff0c3124c55960a18b95433a9c8 : SUCCESS in 1h 43m 09s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2961bfb3b7914f60989196ff0d4e7bf9 : SUCCESS in 29m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4dc88c794974e1a91d35b79019c3634 : SUCCESS in 1h 44m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1cc67525608477895be84d289c14dd5 : SUCCESS in 54m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"e1af9e980793d3117b14f00b07d3bee61bc80ab2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-24 20:23:54.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-21-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-21-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"a795184c56c02b1037b8d9c8775b4717d5c612ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 20:45:13.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_08_2026_20_38_04\nLeave a comment with","accounts_in_message":[],"_revision_number":21},{"id":"f5e935c37f917a24537dda0fc18683a182a3122c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:53:07.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7126d44a354d25b9b94bdde7c1e281\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dff4461483bc4e15a0dd6115ae1a702e : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/474dae8e6fa64f85a2f31d283e88792c : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/12adad22232242f9a2241650afce5b87 : SUCCESS in 18m 20s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ac20f9f47db1cb295d9c75638510526ad94437fe","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-25 03:21:17.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"ad7b025c35f3107e11b2f701c10156739e34c3e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 06:16:19.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5054/.","accounts_in_message":[],"_revision_number":19},{"id":"5fe7574177b3b8ef126a50afd5a677297c66f86d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 08:53:28.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5055/.","accounts_in_message":[],"_revision_number":20},{"id":"27c593802ed05bf888578f9bca0b426f6fbd81dd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 22:13:58.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5060/.","accounts_in_message":[],"_revision_number":21},{"id":"e79a76d6ee51f31f1ef5b8adf979b021cd64a1dc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-27 15:24:33.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"81a695368103e88919b7cc7df4d53ced8ef391b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 15:26:41.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"226291f51d3580993a1fb0739e588c5cfe148cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:16:47.000000000","message":"Patch Set 21: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29be6047e51b457496ad737142ff9ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df60dbcff8fe456ba3f2a198eb2213ab : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2920bd699a5e4022a23b6774177e2eb4 : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2f2c4e19f4047fc9767e17e7137d0f0 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7ce6eb3a874fc1900f3e3690dcaef1 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a769c2c04935435a98f51ae6442c5e1e : SUCCESS in 35m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/582901b1154546b3825236c7d5571cf6 : SUCCESS in 1h 06m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3df845a0fe74d75b445ae2a4a45ed33 : SUCCESS in 1h 01m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fcabec7203483481411f018fb9d661 : SUCCESS in 1h 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5480f70058364b399d8efb9965e60efa : SUCCESS in 5m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c725760d447944858adcda02d193ee1c : FAILURE in 1h 05m 48s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1e6423673d641cfaeb8c9580b8a661a : SUCCESS in 1h 37m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6979643cc484c4bb86beb304daf355c : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":21},{"id":"ded82875beb77bc36bca3551228eb5a83288227a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-27 20:29:07.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"7b0f343dd555dd395d2413bf19c65ba972981b31","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-27 20:40:32.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-21-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-21-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":21},{"id":"b47a1c7746577cd47cbfab9e29345bcda670f3e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-27 21:21:03.000000000","message":"Patch Set 21:\n\nBuild failed.\n\n- cinder-storpool-tempest build/761b3cc9a08c49728c259b6e8b101398 : SUCCESS in 48m 16s\n- cinder-storpool-tempest-iscsi-multipath build/532042304e17412397c5c3dbfe19355f : FAILURE in 50m 22s","accounts_in_message":[],"_revision_number":21},{"id":"a00daf51afbee48fa7977ce1b0cd0d492db6cae4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 21:30:49.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79d2ff7efa242a2afaf1d9c0cc47180\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b5e1ee739bf49b1a90a66740b20a30b : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a603549a61343dbab64c3fdf50b8b19 : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0cbf316df3854cf6ba5a3c59734300e9 : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"197b4248292cafaa6d7b245fe5f0ca88d02ed9d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 23:10:34.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9549abcffe442638c5b93ee9bbd31ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3db169ef79413984bb91d1ec4fceb2 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b327090444a409989b3edcf4e98e1a3 : SUCCESS in 7m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd9a89b29cef4921b18c7c4ad5396142 : SUCCESS in 10m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efd2efba462d457ab5b4a063f3be83bb : SUCCESS in 14m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91803b8c8844acf9eca3b969f7d3cee : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/0289663439de4f158f76ee366eb8e40b : SUCCESS in 1h 08m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2247d9e2a9e40148c59e86f5655789b : SUCCESS in 1h 00m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9574bb43b272498e83ce1054b090ef1e : SUCCESS in 1h 30m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f5a7fc0d1f549a982bfd25b1c8e7991 : SUCCESS in 56m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6801498e19cb420b8d4aa4c4fcb5020c : SUCCESS in 9m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4aa8e828dd714b3e913b25cc48da4e46 : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/22da2df2d324464898af866f75425c84 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60a51ba337be4606befa6865ca0710c1 : SUCCESS in 3m 14s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/19b4c6bbcd7a46069b1bafaac7fb2112 : SUCCESS in 11m 12s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e0d032211d09461e80cf2ebd0fd4c5b6 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ebb567e375b84564a99c7c63beb028b0 : SUCCESS in 4m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6cbdc6f972e43d48ea312bc1eabce78 : SUCCESS in 5m 37s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/aad3fbf8717d4ea18a6223df33794d30 : FAILURE in 5m 38s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2037345987f649099f2bf894c89f939e : SUCCESS in 9m 02s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2ddf42cc92146bbbb33ccd76346f8c3 : SUCCESS in 38m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14ba205b90b74b0a890febf63105d0cc : SUCCESS in 6m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/411ac0e555744f4e90660c1fcd9caf83 : SUCCESS in 1h 11m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/72e516bd788643a5aa8a94c138df6d5d : FAILURE in 56m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f15c1de582f0422cad1691087ae60fbe : SUCCESS in 1h 57m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/285fc2a4edad4f2c93b1af5792042967 : FAILURE in 2h 35m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f79e974cf7a9427890f7b93282ce9cb6 : SUCCESS in 2h 02m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/73bb5563b8a543a495268bf99ca0368c : SUCCESS in 34m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2a16badeed864724b40b14a980c1162b : SUCCESS in 1h 13m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7f1ac62d27842a79e8e37070afbe207 : SUCCESS in 1h 32m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ebfa80f884b4c43b9cdd8a5c64d5035 : SUCCESS in 24m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cba0ce6cbf554d3c9159095c8157dea3 : SUCCESS in 42m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15959af839e44d329e6e94fbfd7fd110 : SUCCESS in 56m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"f8b3403d26304bcd62cc67a910c0a6882312a3d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 23:21:49.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_08_2026_23_14_17\nLeave a comment with","accounts_in_message":[],"_revision_number":21},{"id":"1ed6efa6119278c6f0781e1c11e87aab16a8b560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 00:23:23.000000000","message":"Patch Set 21: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 993363,21","accounts_in_message":[],"_revision_number":21},{"id":"76693b43f43d9140c19761d70dd23aaf896d8ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 02:37:35.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"fc4ffe7373be0464c0eadf4de2501609923a82a6","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 02:41:47.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"f658c60047f54b197017290f1a7315f91a4c3f1d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 02:47:20.000000000","message":"Patch Set 22:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5278/.","accounts_in_message":[],"_revision_number":22},{"id":"0bb6ff90bdf1803fe54bc24fa4e0149e30c20005","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 03:12:36.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2d2e3a927f90440fb6831cc51a943530/","accounts_in_message":[],"_revision_number":22},{"id":"a6f1ddad6f3f6c256e05eafe93a5381c134e8415","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 03:26:15.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2d2e3a927f90440fb6831cc51a943530\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c73be8274fe34784a50d42ecb2cb3471 : SUCCESS in 48m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93657e6845734c48b3e8a0ef89f7ae2e : SUCCESS in 39m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/41a4ccf14aa444c7beb00b0980683f44 : SUCCESS in 47m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/540b7813b4f6405b93ff9014cf07c493 : SUCCESS in 34m 31s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f259015710eb48cfbfc0a0e72c413ffa : FAILURE in 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"259e6500d6b150d6fe86a557ed734993030c0b7c","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 03:27:15.000000000","message":"Patch Set 22:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1057 : SUCCESS in 0h 49m 19s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993363","accounts_in_message":[],"_revision_number":22},{"id":"7d3368eb06f0c1321aa57bdf051f352d7967bbcd","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 03:28:08.000000000","message":"Patch Set 22:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1326 : SUCCESS in 0h 50m 12s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993363","accounts_in_message":[],"_revision_number":22},{"id":"7f9a54c1c2ab9b0e16220a16de8a141e3940baa1","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 03:28:16.000000000","message":"Patch Set 22:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/964 : SUCCESS in 0h 50m 19s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993363","accounts_in_message":[],"_revision_number":22},{"id":"69ac4bc455c3f254f431942eb6f0c93bb9be25ef","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 03:29:16.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7aa72bb0d7b24faf82b2cf627d5c4d9d : SUCCESS in 46m 17s\n- cinder-storpool-tempest-iscsi-multipath build/38ec44e93c73441eb381f0334a2e7d1a : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":22},{"id":"8eba9ac9973c4cf7692f7a950ee6c21056a5a1b7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 03:32:03.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ib07d2cd496cdac490313b69e55b03093b012af9d .","accounts_in_message":[],"_revision_number":22},{"id":"3525346a0504b4e3fa6fc500c9c08056ccc6e1bd","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 03:42:03.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"32c53b9f4665a8d454914de70bedb77e3eb694ba","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 03:47:14.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"161a9a992e0d49c74468c0c399aa51788f035467","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 04:18:22.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/56800d81e5004549af8c46efe2c4381c/","accounts_in_message":[],"_revision_number":22},{"id":"db4322b907a28158cdd17e8f1924185ef9b941d0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 04:35:51.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/56800d81e5004549af8c46efe2c4381c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7f2f6a75c363404d84febc7bc2437104 : SUCCESS in 39m 50s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c26ee0cc570449580d2f55d5168b9bf : SUCCESS in 35m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a289494f16cb41409057b0b80440026d : SUCCESS in 47m 20s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d9500d0333dc4403b2dcbd25914ad87b : SUCCESS in 30m 41s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95906145f2bb4058958b87726d5fe276 : FAILURE in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"2cd5f606e4bdbd4a272efeac3f9e3e1a4e171f35","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 04:43:05.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1189652371da4a34b97cbba0781e40f9 : SUCCESS in 46m 34s\n- cinder-storpool-tempest-iscsi-multipath build/f49664fb3a0140e5bf8b48873b4c0060 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":22},{"id":"69b684ff76ab1692b0af6088343fb567d48952a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 04:54:09.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03e9cfe446a842f98213c10957ad2aef\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98838acfb0284917a59a265419e56fa0 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1970cfee493e46cd9d2fd783ed6fd762 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/509960ace7944d0fae90c54169c1355d : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f3e7e97a939b724879b6fb5dbcf3180579bd3f92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 05:18:20.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07dc661d123e41db8944383bf0aeec90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89a237c120048789d52b5594f5cb75d : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c49e78de81f348ee96b06c28d17f08c2 : FAILURE in 17m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b4482af858b4d35b1f76a4503fa3ffd : SUCCESS in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f780848247b546ddac90d4ea494fe390 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d4200f224f4a649f7e435e1a24f8f7 : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0e1f0183b73a4ddf8061380bfdda3d05 : SUCCESS in 1h 10m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c89157222d342749d4a00c085197a3a : SUCCESS in 47m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4013532913c5422ea0aa56fd63bbc7be : SUCCESS in 1h 28m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcb253b5da6a49288c412d246afa94e9 : SUCCESS in 1h 00m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1282d60c1394c2287320ba88d34c6f1 : SUCCESS in 10m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fb989efa5f4e40d19811db9beb7aded8 : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2b360ddc421e4183b69d2e7019103ced : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b413f1aa2eed4b27ba2ca836d4f72d6b : SUCCESS in 7m 53s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/33fb0b7b66664ed896e7214a13693c04 : FAILURE in 11m 44s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/94a152732b644dc6b3f4448efaf5bddd : FAILURE in 11m 46s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bea80fa1422a4652b6fbfebb4fd2ab4b : SUCCESS in 7m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a40734863f494127b80b29ab4b7cf044 : SUCCESS in 7m 01s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0c7398228ec3475fb7fcca0bb4ecd20a : FAILURE in 6m 00s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/95452b0f06b24a76b37a42aac77ab885 : SUCCESS in 9m 31s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/89fd1be29ce842e9964467e5702e399b : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1e6cd884a9e446e96858f92a7f9a8a8 : SUCCESS in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d903c788e6934b908b4663dacfa18df6 : SUCCESS in 1h 39m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e724d0f0cd1f4c6a838ff41cec48938f : FAILURE in 1h 59m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1a8a745cdd0406ebfe93edf1a4415d6 : SUCCESS in 1h 45m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6682054120e3450c9c47119f42beac72 : FAILURE in 2h 22m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/31100a36bde542678367ee1ebe0127e3 : SUCCESS in 1h 35m 37s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/67dbb25cfccf441e8bdabc8f8da0d450 : SUCCESS in 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d540d526f4ea4a5dab96fa08074afe99 : SUCCESS in 1h 20m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5b5dbbdff7c647d3b8f3b1a323133120 : SUCCESS in 2h 01m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9972aff480344ff6a04b3b00f4035a3f : SUCCESS in 1h 00m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/345b4c55d08548558a8115677470a519 : FAILURE in 1h 39m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2020d47dbe2b48b4824e79465c7dd1ba : SUCCESS in 56m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"84f59a22d4bc98a080ae63c171fb13a00a41f9a9","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 05:24:04.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"20f0673ea5e1add8729d25d887c4249e58f7c90d","author":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"date":"2026-08-28 07:08:49.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"7434f394e2d86be6cc75bfe1b7a1a2709a0d266b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:26:27.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/491e07129bf841e093711a24bb67214c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e460798977504dd1b234aa0ed1620c3b : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60291b2b932d4bd6a583de496ce5c08a : SUCCESS in 20m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/18bfb8de57324466ad27327f67c9785e : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"dbddf165192d213568f051ae3367db71fa56f856","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 07:27:06.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/7da0b1a6b66d4e3589aa8dc7b7b25620/","accounts_in_message":[],"_revision_number":22},{"id":"186bc1bce9528fa5dbf81750fecc1b07c5810f97","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-28 07:37:57.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/22/upstream-check/cinder-tempest-plugin-ontap-fcp/691ba6e/ : FAILURE in 25m 03s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/22/upstream-check/cinder-tempest-plugin-ontap-iscsi/c43ab06/ : SUCCESS in 27m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/63/993363/22/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1b6af84/ : FAILURE in 26m 25s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"cf8378e66f941d8710e7c890fb94e4f45a3a6b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:38:01.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff79d24479e4e07af5e3355d32cb09b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2f997e93d345cf8310c67c9420bf96 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c283511328dd468f965d084f3868cfa7 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ebb8e73151d4b93b95339284847555e : SUCCESS in 10m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0aeb98b25c2a4ead9a11ae8957641a7a : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c132d937744b5491b077c6fe8b67d0 : SUCCESS in 14m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/1954a59f712f431aa33c17c1a327025b : SUCCESS in 33m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5aa7de64cd63413dbe04a58bf3e714b1 : SUCCESS in 1h 05m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87dea1cfb89b421091b88fe0bb6ca2d6 : SUCCESS in 1h 25m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1a8da36cbf8457f9d77bd372221e694 : SUCCESS in 58m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c55af6d6637b4d56b06757356a1975a0 : SUCCESS in 9m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c60f528cc7974c588cba70acf070453a : SUCCESS in 22m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b94c2f6fc7f47b6b52b6c4af9830033 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e8f4d38cee5a4b368b63dac675d15f95 : SUCCESS in 7m 09s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bc00389cdac94e80bab53ed7212d8f94 : FAILURE in 11m 29s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b3f3040fed734d3582cc6803e5b813c3 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f1b6722701e4b1a96d943960369fdac : SUCCESS in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/22aaeeb159f24265a0c2c5c19d7b3d8e : SUCCESS in 5m 45s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/99d5c6017bab45f0904db6d8c0ff11db : FAILURE in 5m 33s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4f609b21023940cf8e03c4282971da62 : SUCCESS in 11m 01s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a530cf494bf49c69ca73dbf1a6c9012 : SUCCESS in 35m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64d16391936b48bbb5a0b5ac6d27ad7c : SUCCESS in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bbf7e760d25e4520b40e36f60717d3a8 : SUCCESS in 1h 43m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3bc68049a724b9da2817f61efbd5bcf : FAILURE in 2h 08m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ef6299719d645f1ac4844344353b5fa : SUCCESS in 1h 15m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e9189753adef484998dd68d2c4f63660 : FAILURE in 1h 58m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/bf5b71957ea54ef7aed531a27e2c8188 : SUCCESS in 1h 59m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a758a6d907824f2eb699c72981c2cf93 : SUCCESS in 33m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5edc5629d2e44497845c569368f772e9 : SUCCESS in 1h 14m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4975e1ea3e214c4396b8a096302d46c9 : SUCCESS in 1h 27m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1064ca4297224b498938ebf8bdb80eac : SUCCESS in 45m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0505d99d39a42bca27b48ac9ff61927 : SUCCESS in 1h 36m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeb954fdcf404833ba8cc0aa8378c327 : SUCCESS in 1h 06m 53s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"3e564ad29cbe741db266a5ab1df533beb7d1f095","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-28 07:55:16.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_08_2026_07_48_10\nLeave a comment with","accounts_in_message":[],"_revision_number":22},{"id":"dc22743c1bab7b2c423701b3bedc45c4a4612eaa","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-28 09:12:27.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"462fcc0954dd911bc714c740aae9bbf4d3a6090e","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-28 09:23:11.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"368ec96ef8741587876bcfc0c93b5eead0f67ef8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 09:37:40.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/7da0b1a6b66d4e3589aa8dc7b7b25620\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e4380514b43d41269c7465618bb8e9b1 : SUCCESS in 44m 53s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9af022455db449c38dbde29539ca0c1e : SUCCESS in 35m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/274a7f7ac9ef42428fdad2a17b9f187b : SUCCESS in 54m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6ca40fee33ce436080eca13fff31531b : SUCCESS in 35m 10s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3af73306f8e4537a9d4c3986d23bc40 : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"4c19cc2f2e7a577b2d0df33ae4ba2ecf50d5eac5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 11:42:11.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5fab6309ce4343ee8e8375599e07cf4a : SUCCESS in 49m 13s\n- cinder-storpool-tempest-iscsi-multipath build/abd9fed9168e4ec7bcdc812b77aa493c : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":22},{"id":"356d862265f84637adc6643a435dbb3c0d8ec8cb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-28 15:47:59.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"bc7bf4819b9bdc1e5f89590c5ebed4fb4abfb32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 15:48:27.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"227f7fb1e79ce4d9da724ac2c121cf614ead6f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:10:35.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1a790e6a2e48b994c1303bc69fe48b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185d91071ca644f1a66045b37316d730 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d378c9f11e084eff9c870e6be0ecfc78 : SUCCESS in 9m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3855bfe1460440b4a8236961226ec558 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc00bfa6ce34ef3a3c0d2c721c9f557 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/5e967fc221334038b478bccd1a2bea33 : SUCCESS in 53m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b122302c350543c88f4685d79df34e3c : SUCCESS in 52m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5aa8f6331fc4022a297739749c37403 : SUCCESS in 55m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe1f239ac5b342dbb61d49f35febc5d1 : SUCCESS in 48m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ac41c61fdb94f47916933e618bcf674 : SUCCESS in 4m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3d4090950b47490fba0eaf7248e436cb : SUCCESS in 1h 16m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/483eb47a027944339ce94ef85bd5c0d1 : SUCCESS in 46m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e583e92f4d804f3f968d9539b5215800 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":22},{"id":"e8c4057370adc2d7ae0565b637d1bc31ddbdf2db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:10:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"fb7ff6857c593a2908028f8d7846461e974a8bed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 17:11:43.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9082f603e52f4b90866a58d76b0696d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a365d99dfc42a59ae747e5f5fad4d6 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e442f7a442ec473bb5bda9fced85d660 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":22},{"id":"fd990694bb17b3f83f20c5c12c92b54bc1b37df7","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 09:27:58.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-22-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-63-993363-22-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"2278582d2586d012da2fe602a67416858cfad778","revisions":{"f86a6c8d5076ef6a861cdec8611f6b8dbaf9f717":{"kind":"REWORK","_number":1,"created":"2026-06-15 14:01:40.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/1"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-15 13:42:18.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-15 14:00:09.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations; additional items will\nbe added in subsequent revisions of this change.\n\nThis revision adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f86a6c8d5076ef6a861cdec8611f6b8dbaf9f717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f86a6c8d5076ef6a861cdec8611f6b8dbaf9f717"}]},"branch":"refs/heads/master"},"9c6fc3be4de09cb85eb4a73cd479b9ee523b52d3":{"kind":"REWORK","_number":2,"created":"2026-06-17 10:09:17.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/2"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-15 13:42:18.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-17 10:06:04.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations; additional items will\nbe added in subsequent revisions of this change.\n\nThis revision adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c6fc3be4de09cb85eb4a73cd479b9ee523b52d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c6fc3be4de09cb85eb4a73cd479b9ee523b52d3"}]},"branch":"refs/heads/master"},"4d7717e0f3d74a80dc117931dcb6cd8164f14001":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-18 06:07:19.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/3"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-15 13:42:18.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-18 06:03:25.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d7717e0f3d74a80dc117931dcb6cd8164f14001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d7717e0f3d74a80dc117931dcb6cd8164f14001"}]},"branch":"refs/heads/master"},"74e703b7e2f8c82281547477f03f36ea4536a589":{"kind":"REWORK","_number":4,"created":"2026-06-22 14:16:45.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/4"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-15 13:42:18.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-22 14:15:32.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74e703b7e2f8c82281547477f03f36ea4536a589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/74e703b7e2f8c82281547477f03f36ea4536a589"}]},"branch":"refs/heads/master"},"f72d004a8c991cd9ab29bdfdd4c5aa5dfda015f0":{"kind":"REWORK","_number":5,"created":"2026-06-24 17:50:40.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/5"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f72d004a8c991cd9ab29bdfdd4c5aa5dfda015f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f72d004a8c991cd9ab29bdfdd4c5aa5dfda015f0"}]},"branch":"refs/heads/master"},"c9e71e284faef0167e92b0838f3982894578cb98":{"kind":"REWORK","_number":6,"created":"2026-07-07 12:33:17.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/6"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-07 11:13:47.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9e71e284faef0167e92b0838f3982894578cb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9e71e284faef0167e92b0838f3982894578cb98"}]},"branch":"refs/heads/master"},"2ec932f36fc9bda12f7140a1c12528557aa29f6b":{"kind":"REWORK","_number":7,"created":"2026-07-13 14:14:39.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/7"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-13 14:12:00.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ec932f36fc9bda12f7140a1c12528557aa29f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ec932f36fc9bda12f7140a1c12528557aa29f6b"}]},"branch":"refs/heads/master"},"32ed11fde7c8b97cdefb7d359b2421cfc9ac3115":{"kind":"REWORK","_number":8,"created":"2026-07-13 14:41:22.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/8"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-13 14:37:09.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32ed11fde7c8b97cdefb7d359b2421cfc9ac3115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/32ed11fde7c8b97cdefb7d359b2421cfc9ac3115"}]},"branch":"refs/heads/master"},"bf34e83392ab88a471ebce8c511ecad642cc474e":{"kind":"REWORK","_number":9,"created":"2026-07-13 15:10:38.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/9"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-13 15:10:21.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf34e83392ab88a471ebce8c511ecad642cc474e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bf34e83392ab88a471ebce8c511ecad642cc474e"}]},"branch":"refs/heads/master"},"4bf52f2234ef82cb74dcd6fa9bc495626e2960fd":{"kind":"NO_CHANGE","_number":10,"created":"2026-07-13 15:11:32.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/10"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-13 15:11:16.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bf52f2234ef82cb74dcd6fa9bc495626e2960fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bf52f2234ef82cb74dcd6fa9bc495626e2960fd"}]},"branch":"refs/heads/master"},"fc56a638a36cdcff61b462f81e90c3574d5c0413":{"kind":"REWORK","_number":11,"created":"2026-07-14 12:43:02.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/11"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-14 12:42:47.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid}.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc56a638a36cdcff61b462f81e90c3574d5c0413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fc56a638a36cdcff61b462f81e90c3574d5c0413"}]},"branch":"refs/heads/master"},"55460c6ad6e1494f2c5470be6b0728cff1254357":{"kind":"REWORK","_number":12,"created":"2026-07-21 10:43:41.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/12"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-21 10:43:17.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nTwo additional bugs are fixed as part of this same REST parity work:\n\n* create_group_snapshot could clone from live volume data instead of\n  the crash-consistent group snapshot, because source_snapshot was\n  silently dropped while building the clone source path in\n  _clone_backing_file_for_volume. The parameter is now forwarded to\n  clone_file so the clone reflects the group snapshot.\n\n* clone_file and clone_lun could fail to clone from a named snapshot on\n  Storage VMs where snapshot directory access is disabled. Both now\n  check the snapshot_directory_access_enabled flag and fall back to\n  the private CLI clone path with an explicit snapshot name when\n  needed.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55460c6ad6e1494f2c5470be6b0728cff1254357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/55460c6ad6e1494f2c5470be6b0728cff1254357"}]},"branch":"refs/heads/master"},"3aefe02f61ffdfb39cad70a4f133658f99f9ac77":{"kind":"REWORK","_number":13,"created":"2026-07-21 10:46:52.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/13"}}},"commit":{"parents":[{"commit":"7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1","subject":"Merge \"Fix missing volume delete notifications on group delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea3e9c10eb32289444c8bf8ce62b4a0d5ace1e1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-21 10:46:27.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nTwo additional bugs are fixed as part of this same REST parity work:\n\n* create_group_snapshot could clone from live volume data instead of\n  the crash-consistent group snapshot, because source_snapshot was\n  silently dropped while building the clone source path in\n  _clone_backing_file_for_volume. The parameter is now forwarded to\n  clone_file so the clone reflects the group snapshot.\n\n* clone_file and clone_lun could fail to clone from a named snapshot on\n  Storage VMs where snapshot directory access is disabled. Both now\n  check the snapshot_directory_access_enabled flag and fall back to\n  the private CLI clone path with an explicit snapshot name when\n  needed.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aefe02f61ffdfb39cad70a4f133658f99f9ac77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3aefe02f61ffdfb39cad70a4f133658f99f9ac77"}]},"branch":"refs/heads/master"},"631306ac330e52ce2714106cf1ef48ede7f6e097":{"kind":"REWORK","_number":14,"created":"2026-07-28 12:52:01.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/14"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-28 11:47:32.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* get_cluster_peers -- lists cluster peers via GET /api/cluster/peers\n  with optional name filter; returns ZAPI-compatible dicts.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nTwo additional bugs are fixed as part of this same REST parity work:\n\n* create_group_snapshot could clone from live volume data instead of\n  the crash-consistent group snapshot, because source_snapshot was\n  silently dropped while building the clone source path in\n  _clone_backing_file_for_volume. The parameter is now forwarded to\n  clone_file so the clone reflects the group snapshot.\n\n* clone_file and clone_lun could fail to clone from a named snapshot on\n  Storage VMs where snapshot directory access is disabled. Both now\n  check the snapshot_directory_access_enabled flag and fall back to\n  the private CLI clone path with an explicit snapshot name when\n  needed.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/631306ac330e52ce2714106cf1ef48ede7f6e097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/631306ac330e52ce2714106cf1ef48ede7f6e097"}]},"branch":"refs/heads/master"},"42a20a4929892ee281ea72e8c1f76b86139f05c7":{"kind":"REWORK","_number":15,"created":"2026-07-30 11:26:21.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/15"}}},"commit":{"parents":[{"commit":"8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bb9a7f7d1fb0fad7cb26d2aece58014d60e6cd4"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-07-30 11:25:06.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVM\u0027s FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshot\u0027s owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nTwo additional bugs are fixed as part of this same REST parity work:\n\n* create_group_snapshot could clone from live volume data instead of\n  the crash-consistent group snapshot, because source_snapshot was\n  silently dropped while building the clone source path in\n  _clone_backing_file_for_volume. The parameter is now forwarded to\n  clone_file so the clone reflects the group snapshot.\n\n* clone_file and clone_lun could fail to clone from a named snapshot on\n  Storage VMs where snapshot directory access is disabled. Both now\n  check the snapshot_directory_access_enabled flag and fall back to\n  the private CLI clone path with an explicit snapshot name when\n  needed.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42a20a4929892ee281ea72e8c1f76b86139f05c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42a20a4929892ee281ea72e8c1f76b86139f05c7"}]},"branch":"refs/heads/master"},"5039dc6e66aff892421cb1a4e49674630018cb9d":{"kind":"REWORK","_number":16,"created":"2026-08-17 09:44:44.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/16"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-17 09:21:22.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5039dc6e66aff892421cb1a4e49674630018cb9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5039dc6e66aff892421cb1a4e49674630018cb9d"}]},"branch":"refs/heads/master"},"cc25f63d0742ec752f13d91c3dc622163ae67664":{"kind":"REWORK","_number":17,"created":"2026-08-17 10:12:15.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/17"}}},"commit":{"parents":[{"commit":"b5b763129e2bde5077c0cf3a5eb434021abaa6e0","subject":"Merge \"Centralize cooperative yield for eventlet removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b5b763129e2bde5077c0cf3a5eb434021abaa6e0"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-17 10:03:53.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc25f63d0742ec752f13d91c3dc622163ae67664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc25f63d0742ec752f13d91c3dc622163ae67664"}]},"branch":"refs/heads/master"},"33bf097899a42c88ec40a5a101b09533cadd9aa3":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-21 15:43:38.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/18"}}},"commit":{"parents":[{"commit":"91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1","subject":"Merge \"Add test coverage for utils.tpool_wrap()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-21 15:38:21.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also fixes the NFS file-copy cancel cleanup path: the ZAPI\ndestroy_file_copy sent ``file-index`` as an integer, which ONTAP\nrejects, so the cancel always raised; _cancel_file_copy now passes it\nas a string and always removes the partially-copied destination file,\ntreating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33bf097899a42c88ec40a5a101b09533cadd9aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33bf097899a42c88ec40a5a101b09533cadd9aa3"}]},"branch":"refs/heads/master"},"3310c1a327ee6c31ac174dea3cdda1e489f8fccb":{"kind":"REWORK","_number":19,"created":"2026-08-24 15:12:53.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/19"}}},"commit":{"parents":[{"commit":"91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1","subject":"Merge \"Add test coverage for utils.tpool_wrap()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-24 14:59:55.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also updates the NFS file-copy cancel cleanup path\n(_cancel_file_copy in nfs_cmode.py): the new REST destroy_file_copy\nonly stops the job and leaves the partial file behind, unlike ZAPI\u0027s\nfile-copy-destroy which removes it on success. _cancel_file_copy now\nunconditionally removes the partially-copied destination file for\nboth clients, treating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3310c1a327ee6c31ac174dea3cdda1e489f8fccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3310c1a327ee6c31ac174dea3cdda1e489f8fccb"}]},"branch":"refs/heads/master"},"f6bdd458c03b3ce72ae4db3fa1953a8c7f803ad0":{"kind":"REWORK","_number":20,"created":"2026-08-24 15:30:08.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/20"}}},"commit":{"parents":[{"commit":"91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1","subject":"Merge \"Add test coverage for utils.tpool_wrap()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91fb9f5e7f9bf9b9b67cc6faed22ac3578c182a1"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-24 15:26:51.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also updates the NFS file-copy cancel cleanup path\n(_cancel_file_copy in nfs_cmode.py): the new REST destroy_file_copy\nonly stops the job and leaves the partial file behind, unlike ZAPI\u0027s\nfile-copy-destroy which removes it on success. _cancel_file_copy now\nunconditionally removes the partially-copied destination file for\nboth clients, treating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6bdd458c03b3ce72ae4db3fa1953a8c7f803ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f6bdd458c03b3ce72ae4db3fa1953a8c7f803ad0"}]},"branch":"refs/heads/master"},"f0eb64a731ea25f4a67f24a5bb9886eb24b7a86d":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-08-24 17:45:16.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/21"}}},"commit":{"parents":[{"commit":"ca4708870cf0516274263c6c45b44705bd593cc0","subject":"Merge \"NetApp:Fix to handle error during consistency group expand operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca4708870cf0516274263c6c45b44705bd593cc0"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-24 15:36:52.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also updates the NFS file-copy cancel cleanup path\n(_cancel_file_copy in nfs_cmode.py): the new REST destroy_file_copy\nonly stops the job and leaves the partial file behind, unlike ZAPI\u0027s\nfile-copy-destroy which removes it on success. _cancel_file_copy now\nunconditionally removes the partially-copied destination file for\nboth clients, treating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0eb64a731ea25f4a67f24a5bb9886eb24b7a86d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0eb64a731ea25f4a67f24a5bb9886eb24b7a86d"}]},"branch":"refs/heads/master"},"2278582d2586d012da2fe602a67416858cfad778":{"kind":"REWORK","_number":22,"created":"2026-08-28 02:37:35.000000000","uploader":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"ref":"refs/changes/63/993363/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/63/993363/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/63/993363/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/63/993363/22"}}},"commit":{"parents":[{"commit":"7b02a990b89e1230e53ae25c4235b28adaf6fcab","subject":"Merge \"NetApp ONTAP: fix ZAPI destroy_file_copy file-index type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b02a990b89e1230e53ae25c4235b28adaf6fcab"}]}],"author":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-06-24 11:44:32.000000000","tz":-240},"committer":{"name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","date":"2026-08-28 02:30:07.000000000","tz":-240},"subject":"NetApp ONTAP: begin closing ZAPI-to-REST gaps","message":"NetApp ONTAP: begin closing ZAPI-to-REST gaps\n\nSeveral primitives implemented on the ZAPI client are missing on the\nREST client and silently fall through to ZAPI via __getattr__, breaking\nparity when operators run with netapp_use_legacy_client\u003dFalse. This is\nthe first batch of native REST implementations.\n\nThis patch adds:\n\n* destroy_file_copy -- cancels an in-progress file copy via\n  PATCH /api/cluster/jobs/{uuid} and removes the partially-copied\n  destination file.\n\n* set_lun_qos_policy_group -- assigns a QoS policy group to a LUN via\n  PATCH /api/storage/luns/{uuid} with ``qos_policy.name``. Unlike the\n  ZAPI client, the REST client uses a single field for both adaptive\n  and non-adaptive policies.\n\n* get_snapshot -- looks up a single snapshot via\n  GET /api/storage/volumes/{volume.uuid}/snapshots filtered by name and\n  returns a ZAPI-compatible dict (name, volume, busy, owners). A\n  wait_for_busy_snapshot helper polls it until the snapshot is no longer\n  busy.\n\n* rename_snapshot / mark_snapshot_for_deletion -- renames a backing\n  snapshot to the ``deleted_cinder_\u003cname\u003e`` GC prefix by resolving the\n  snapshot uuid via GET and issuing a PATCH; preserves snapshot\n  identity.\n\n* get_snapshots_marked_for_deletion -- lists ``deleted_cinder_*``\n  snapshots across the SVMs FlexVols for the periodic reaper. REST has\n  no server-side busy filter, so busy snapshots are skipped client-side\n  via the snapshots owners[] array.\n\n* get_file_usage -- retrieves the unique-bytes count of a file via\n  GET /api/private/cli/volume/file/show-disk-usage. Used by the NFS\n  image-cache cleanup path. No public REST equivalent exists for the\n  ZAPI file-usage-get unique-bytes attribute; the private CLI endpoint\n  is a short-term workaround.\n\n* get_clone_split_info -- fetches unsplit clone status via\n  GET /api/storage/file/clone/split-status. Maps REST fields\n  unsplit_size and pending_splits to ZAPI-compatible output keys.\n\n* get_flexvol_dedupe_used_percent -- combines get_flexvol_dedupe_info\n  and get_clone_split_info using the same formula as the ZAPI baseline:\n  100 * (logical_data_size + unsplit_size) / logical_data_limit.\n  Guarded by the CLONE_SPLIT_STATUS feature flag (ONTAP 8.3+).\n\n* create_cg_snapshot -- implements crash-consistent group snapshots\n  natively in the REST client. Creates an ephemeral ONTAP consistency\n  group, posts a CG snapshot, then deletes the CG. The per-FlexVol\n  snapshot persists for cloning and is deleted after all clones\n  complete, mirroring the ZAPI cg-start/cg-commit behaviour.\n  _start_cg_snapshot and _commit_cg_snapshot are blocked on the REST\n  client via __getattr__ to prevent silent fallback.\n\nget_cluster_peers ZAPI implementation removed; it was dead code.\n\nIt also updates the NFS file-copy cancel cleanup path\n(_cancel_file_copy in nfs_cmode.py): the new REST destroy_file_copy\nonly stops the job and leaves the partial file behind, unlike ZAPI\u0027s\nfile-copy-destroy which removes it on success. _cancel_file_copy now\nunconditionally removes the partially-copied destination file for\nboth clients, treating an already-absent file as success.\n\nChange-Id: Ib07d2cd496cdac490313b69e55b03093b012af9d\nSigned-off-by: Saikumar Pulluri \u003csaikumar1016@gmail.com\u003e\nRelated: https://etherpad.opendev.org/p/cinder-netapp-zapi-to-rest-gaps\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2278582d2586d012da2fe602a67416858cfad778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2278582d2586d012da2fe602a67416858cfad778"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
