)]}'
{"id":"openstack%2Fcinder~993949","triplet_id":"openstack%2Fcinder~master~Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-13 14:24:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-13 19:18:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0","subject":"Remove unused workers param from service.serve()","status":"MERGED","created":"2026-06-18 14:51:20.000000000","updated":"2026-08-14 00:05:04.000000000","submitted":"2026-08-13 19:18:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993949","meta_rev_id":"53ccf45a78a1187c66a78647517242921e86e819","_number":993949,"virtual_id_number":993949,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 19:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":2,"date":"2026-08-13 17:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":2,"date":"2026-08-13 14:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":1,"date":"2026-08-13 17:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 15:08:09.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-18 15:53:12.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-18 16:27:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-18 16:37:56.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-18 16:43:19.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-18 17:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 17:27:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-18 19:58:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-18 20:39:27.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-20 22:47:13.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-13 14:21:52.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-08-13 17:35:25.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-08-13 19:14:45.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-13 19:20:05.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-13 23:29:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"5267c5c47769e3dda473bc9e3b7f0f564d72406b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-06-18 14:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8fb35b50ffb8c3c0c859f196bbec6238317f0db","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 15:08:09.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/504 : FAILURE in 0h 13m 34s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"771b9d72667bd54b83ec184e466d72b98ff3d79d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 15:35:05.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/236 : SUCCESS in 0h 39m 49s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"ed96db90f3cc934cb14893cdb2107527ab745dc8","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 15:46:13.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/150 : SUCCESS in 0h 42m 43s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"32051515534df1c60805196f8134ea1bc2dd6a7a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 15:53:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0 .","accounts_in_message":[],"_revision_number":1},{"id":"1a2f490f58e29eee32fdd53265c87a682d5f69d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-18 16:27:37.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3a4d6d7ce17b44bbb38eea9c509f86fd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/338a18177aa64b4a89879a3d6ca2cef7 : FAILURE in 44m 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/21ccbb8d9383411c91540a361589e46c : FAILURE in 32m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d1f1a32189146e3919e94b4dac5f7ad : FAILURE in 1h 35m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/13f31b8064a74b3085996ff35ca74628 : FAILURE in 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8dd463584027eb71444b943108ca9c7e412459d2","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-18 16:37:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/f64664ecdfb84170ad83e6248209d4ed : FAILURE in 24m 51s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"3e8fdba33d0267f8f8c63d908842dc0fe0193a98","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-18 16:43:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/5a84dc1f8262487ebf96da6a646bec9f : SUCCESS in 34m 54s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"e03fad03e05bbcff476f30a9523baab5ab88664c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:27:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81424467f2914ef48a5c2d880a85bd7c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4730198615714b14890682d60745c99c : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e476d5a5f88149a68d3e975aa908bb6b : SUCCESS in 8m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4350ff7e63e4a808b6d81041162e816 : SUCCESS in 12m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3579abc71f148609c618805918ab08c : TIMED_OUT in 42m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc72990123e843778cea8e58b8431531 : SUCCESS in 17m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/4b8dc002b532411f85d1da021a7ec0f7 : SUCCESS in 33m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3747f2991f2448aea155ccf4d7619678 : SUCCESS in 1h 02m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9505c441a804218b31d09f4c3194ae8 : SUCCESS in 1h 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/072f93405bad45ee8aa9f9733240bf20 : SUCCESS in 1h 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecbec8c76fcf422b8870b82b9c0471ae : SUCCESS in 9m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a519c747dea44e78ae54b4e4d1f4b11 : TIMED_OUT in 41m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/addcceb42da04f8093d3cd293fee16fe : SUCCESS in 6m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8b9838269000436895259e83eaa099e2 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ed889d3de27846a69a441a87242d30e0 : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f2281e094bfb4c5194f029d8cda8d3f4 : SUCCESS in 8m 13s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/10f2d68669c0479288ed25e00beb0da4 : SUCCESS in 9m 30s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1631855f381044d486aef7d4328b12b2 : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/44a9b36b9fd94669a1352f3002d35518 : SUCCESS in 8m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e886fdb81cfc4180bcfa940b6fcccc2e : SUCCESS in 1h 09m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b5c35446ae934615bd4cdc4309e4f75c : FAILURE in 2h 27m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0978bea2f2fd47ec91151703a7d480ae : SUCCESS in 1h 52m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9958bc539079444a8cd8ab3492513c67 : SUCCESS in 1h 55m 11s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a3ffe2e27199479a8ac610b035c20d46 : SUCCESS in 22m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/40d1cad0eba4429884d2b4078b953681 : SUCCESS in 1h 19m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94cb596576554f0ab8fd1a4357fa388e : SUCCESS in 1h 48m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c4d6f1ff037c4587a7a84ad0a8780d3e : SUCCESS in 32m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5abc331a38f240088db68f668ff964bb : SUCCESS in 53m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/086b4869891349b99f59ed382ecaf547 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a0a2de8aef94769f8977e225334ea2754094a666","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-18 17:38:41.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_06_2026_17_31_02\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"0e3e50d63a60896b8326cf6d4db4aa628d30a114","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 18:06:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7883691e218a47a48771e25aeb4cc785\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b27ebfb7efc47e9bc88b94ef6a189a0 : FAILURE in 21m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29e1513432394b92a08590527386318d : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/013065ff71ff4e198ef1777fcfc9be51 : TIMED_OUT in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"846c0349554eab806a57ab08e5ba019cc3493ca1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-18 19:58:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c29f40e09d79458bb1c6dc98f289721e : FAILURE in 31m 13s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993949\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993949","accounts_in_message":[],"_revision_number":1},{"id":"4c9dd15774e1edbf244bce11ca5af1f68204bc23","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-18 20:39:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/f37f6bb81c8d498e86cb994c2bcd6287 : FAILURE in 52m 01s\n- cinder-storpool-tempest-iscsi-multipath build/e33004bf92eb40a49415d0289ffbe09c : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7c47142b4f7991f5f5590e5fead8139facedb61a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-20 22:47:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-993949-1-fc : SUCCESS in 59m 41s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-993949-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e626f15d2db9b73ee50789776242869f9377703d","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 14:21:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3912029c792bdc3bc4a3e6cd381e455cba99f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-08-13 14:23:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"770dfeeaed165613dc43a1bdcb0be657eee8e8ff","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-13 14:24:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"702664dde63d7e4f8a56c052cb62055ab580a790","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 15:24:47.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0 .","accounts_in_message":[],"_revision_number":2},{"id":"ee3df1dc87888ce5b07e9788ad2c050d77c59798","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 15:46:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5b1cb6b272347b6a26b7819b8b5760e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da974e5fd6fb459dbfd63d4464bf2003 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d57d973d60cd4c59983d36e830c5e050 : SUCCESS in 16m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2fd69b04968c471aba286a66768f314a : SUCCESS in 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb57f802edc1a84be8098c52a69750f1037cee6e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 16:51:21.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/c7743f0f6c2a48ccb1e24628eaa80050/","accounts_in_message":[],"_revision_number":2},{"id":"c9c86130219931be51030814304d8aaa5dcffe1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 16:55:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83dab92799148039a2019757a12604c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67053b55cccd4b108e29713517a0857a : SUCCESS in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc7689a0a2624aa49fef7e3b03b9759b : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eec8dc5eb33b4802b289cdf997a1f65f : SUCCESS in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/965f365736fb41abb6e4e7e1f06d9fb1 : SUCCESS in 15m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5311e8e63308406a86f55b209221f662 : SUCCESS in 14m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/19eca59e3cd241588a05ad0bf707d443 : SUCCESS in 52m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7dfc956af17b430189a45b82bb32ea98 : SUCCESS in 48m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2da8b397cc704e2996a715eb3d3450d9 : SUCCESS in 1h 28m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65bdf286ce7942fc8fa9157a93ddd344 : SUCCESS in 1h 06m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab6bc75bacf457aa4cafa81ff069fc1 : SUCCESS in 9m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b73f1d5a4b8e4eee93f73826950b1ba9 : SUCCESS in 26m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b3d03616015e4572870b07b7afbdc74a : SUCCESS in 5m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2dda6f5dd6d14739997586dea5ba3bda : SUCCESS in 8m 00s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/7ba6d56dfffa45b287048532bd055a01 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8a1ff5609fd34720b65244db300f02cf : SUCCESS in 7m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/444a1ea74b134b508a12e466cfa773a7 : SUCCESS in 6m 52s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e6f7cce0e9464d8091c6f441d43eae68 : SUCCESS in 10m 27s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b3bd23d1a4d949769e192341b3cef68e : FAILURE in 28m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9743a919ff5847d480cc8e98ca5122e0 : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/957c69ff904c4625956a8fc477c24b17 : SUCCESS in 1h 02m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6df992553cb0462682d027358ad9f697 : FAILURE in 2h 19m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/23a2a5cc084f42c594b0c0883dc2b664 : SUCCESS in 1h 40m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f0d361b29d05450ba6fe0b77a9584afd : FAILURE in 2h 24m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/be53ad95d20f4abbb812f48c8d0ff3da : FAILURE in 2h 04m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/299a6ab59fbd47218911d81576534342 : SUCCESS in 44m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/57301beae7414aa1bb528963224ca5bb : SUCCESS in 1h 06m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08d06155da474b55940fcf8f62381f73 : SUCCESS in 1h 42m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a924d4d024fc4c9a8f736eabebb5977b : SUCCESS in 43m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d1421b0513a4c14a7b6a291443cb5a8 : SUCCESS in 41m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fdab05753c1445caec4cdff40c94608 : SUCCESS in 40m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"fc6b74de16a05ea84a70577a829114bab0f91513","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-13 17:35:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1049ee3f1b1304c3c6a6c1cab8bfb63d2a740db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:35:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3bc766f8bb5402b090915c5061610f59e90833e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 18:03:17.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2026_17_56_30\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"2b163c4a135eec4103cb4ac95d3ce0fb994709d6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 18:44:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c7743f0f6c2a48ccb1e24628eaa80050\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/79e24a85e5e943de9f2f5c1b8b98e86a : SUCCESS in 45m 01s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cab7f994f4e4efa91867a40e2c28c98 : SUCCESS in 35m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/595d860bb2e94e6bb93425ed081e40e2 : SUCCESS in 47m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7dddd7b541ab426d9c726d6459416e75 : SUCCESS in 35m 35s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1d1a17bb96d74eb2b63afaf7eccbbcb5 : FAILURE in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c5526d98d20f0482fd4420205d00a9004214fb5","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 19:14:45.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/808 : FAILURE in 0h 38m 32s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993949","accounts_in_message":[],"_revision_number":2},{"id":"56cd4cc8fd38104dda4faab4c65be1b23c0d3ce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d33be0256405400d8336b576ea53de7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f3d18cd3d046ab8b0c07db8e95509f : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/735a04c8385c4bb79ca63ce4e12979c4 : SUCCESS in 7m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87f4e9981e304d3191f320dcae528ab7 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64148ac9347c4e439509a131d9795e64 : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f9195b01356b46aea392ce866b5b5863 : SUCCESS in 51m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/237de68aa0ae40b08863a94a1691bd3a : SUCCESS in 33m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc416366241f40af9397f85509f9a3de : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a74813c4f84a4931b77e4a1b845a7a59 : SUCCESS in 1h 01m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fac7868f01994129beb84be27d845c33 : SUCCESS in 7m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9cb2e078431344299177571f33bdeb05 : SUCCESS in 1h 15m 25s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51bd7bd231204f7d9d6e8d192cd03931 : SUCCESS in 1h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2997e0a2ff9c45f690970ecf5f0a3bba : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":2},{"id":"335d3da9926530cb83bb46e9413f91a101f11399","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:18:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"76146d6f3d1856acf8c9d076d7403ba2e2b8b8f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:20:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fbccefdc35b49ff8d8cfe9ed7cdb3b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f755391d8446a1bdfc67f0099ee575 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2},{"id":"5b151c261a1a62cb0a82eb9367872a738683dc8a","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 19:20:05.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/902 : SUCCESS in 0h 47m 12s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993949","accounts_in_message":[],"_revision_number":2},{"id":"27a435e5ebc8ceb0fe99957a60cdf2d3d41bef4d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 19:22:53.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1169 : SUCCESS in 0h 47m 38s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993949","accounts_in_message":[],"_revision_number":2},{"id":"4579791125f0d930d081e0644d01de47f229c166","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-13 20:38:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/daafb7e31a02444198c7bb9837b8d578 : SUCCESS in 45m 43s\n- cinder-storpool-tempest-iscsi-multipath build/54565adf23ab4614979969960a07cf9f : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1dc1b03c517b5742f3ba37b073ee17a314615d73","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-13 23:29:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4939/.","accounts_in_message":[],"_revision_number":2},{"id":"9ee820e1a3bcb4074f3a312c61b21dd6b6139003","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-13 23:32:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-993949-2-fc : FAILURE in 14m 9s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-49-993949-2-iscsi : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":2},{"id":"53ccf45a78a1187c66a78647517242921e86e819","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 00:05:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5115/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9721592f32f56726b983e30585a8c04f737d8c8a","revisions":{"c0fe27336bd0148d7f8f97487a28df382d2e79b8":{"kind":"REWORK","_number":1,"created":"2026-06-18 14:51:20.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/49/993949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/993949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/993949/1"}}},"commit":{"parents":[{"commit":"1bfeea7298f3ffc27a71a7cfa4815d932e32429c","subject":"Merge \"db: Remove unnecessary sqlalchemy abstraction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1bfeea7298f3ffc27a71a7cfa4815d932e32429c"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-06-18 14:27:23.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-06-18 14:28:55.000000000","tz":330},"subject":"Remove unused workers param from service.serve()","message":"Remove unused workers param from service.serve()\n\nworker parameter is not used by any caller in Cinder codebase\n\nChange-Id: Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0fe27336bd0148d7f8f97487a28df382d2e79b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c0fe27336bd0148d7f8f97487a28df382d2e79b8"}]},"branch":"refs/heads/master"},"9721592f32f56726b983e30585a8c04f737d8c8a":{"kind":"REWORK","_number":2,"created":"2026-08-13 14:23:19.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/49/993949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/49/993949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/49/993949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/49/993949/2"}}},"commit":{"parents":[{"commit":"911497601b299bcb679f454fe9f3fb60e66ada94","subject":"Merge \"Run Cinder\u0027s Backup service in native threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/911497601b299bcb679f454fe9f3fb60e66ada94"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-06-18 14:27:23.000000000","tz":330},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2026-08-13 14:22:57.000000000","tz":330},"subject":"Remove unused workers param from service.serve()","message":"Remove unused workers param from service.serve()\n\nworker parameter is not used by any caller in Cinder codebase\n\nChange-Id: Ic7b4ececa8e0c5b5eaee5aaffc167dca6a8743a0\nSigned-off-by: Amit Uniyal \u003cauniyal@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9721592f32f56726b983e30585a8c04f737d8c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9721592f32f56726b983e30585a8c04f737d8c8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
