)]}'
{"id":"openstack%2Fcinder~993979","triplet_id":"openstack%2Fcinder~master~I77291902e477ed14180dbe4692edde6717bbb791","project":"openstack/cinder","branch":"master","topic":"bug/2166187","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-29 14:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2026-09-25 11:27:16.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-09-29 16:07:48.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-09-29 16:07:48.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-09-29 16:07:48.000000000","reason":"Change was submitted"},"38286":{"account":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"last_update":"2026-09-29 16:07:48.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-09-29 05:51:23.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I77291902e477ed14180dbe4692edde6717bbb791","subject":"Dell PowerFlex: Fix image cache vTree size counting","status":"MERGED","created":"2026-06-18 17:44:18.000000000","updated":"2026-09-29 16:10:41.000000000","submitted":"2026-09-29 16:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":82,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993979-bug/2166187","meta_rev_id":"0be24957641b3e821f9ea94c38f20977cd09ef2d","_number":993979,"virtual_id_number":993979,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 16:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-29 14:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-29 05:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-09-29 14:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 18:16:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-18 18:25:33.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-18 18:30:24.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-18 18:32:33.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-18 19:19:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-18 19:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-18 20:18:50.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-06-18 20:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 22:15:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-19 00:39:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-19 10:12:43.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-06-19 13:16:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-19 20:23:31.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-06-22 06:47:19.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-06-22 06:47:19.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-06-28 14:09:48.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2026-08-25 09:47:12.000000000","updated_by":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"reviewer":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"state":"CC"},{"updated":"2026-08-25 09:56:43.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-08-25 10:35:02.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-08-25 10:35:19.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-28 11:36:00.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2026-08-31 03:45:21.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"1d7c3f95fec0566ef1a8698bf0d4da91733e57e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-18 17:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc1ffd5378a64c28c696f78a5e3ca00952d91edd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 18:16:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":1},{"id":"1760a8d6dd92488671bc83bf5727114b27b6846b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 18:25:33.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/153 : SUCCESS in 0h 40m 52s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"a4503d26911dda4743e186d3f11f1168e1d253df","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 18:25:46.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/507 : SUCCESS in 0h 41m 8s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"7c0308f19215dc1bf38137baf482e16726cd0ad1","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-18 18:27:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62558f13ea34ee65224fcde1cccddae3f65426a3","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-18 18:28:08.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/239 : SUCCESS in 0h 43m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"2cde4652e663f1dca5ac0a39c900bd1c3b0d59e8","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-18 18:30:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-nvmetcp-driver build/9f033d3280394bd4972c9b81eac2de5c : SUCCESS in 45m 42s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"a79ba8dd232d8e1406e85c0102a7de18f09e8933","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-18 18:32:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/66fae138607e40e59ded5c9f30aecfd0 : SUCCESS in 34m 37s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"b41d6ff01b42dbbcfbcd3f82ee0a61cbf45ea955","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-18 19:19:17.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e4f5c24e224d430d85b004878222d40c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/bf5a067f89124f389f64ce5f22480b9a : FAILURE in 39m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c74623494d0b476598b51aeceb30689a : FAILURE in 36m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c2e91f5e1bd42dd8c9a01e62a1389c6 : FAILURE in 1h 34m 28s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6878ee009570499cb2703a426b09b005 : FAILURE in 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2cbc51c1c47b11cc08fca9bc080d992fe825d3a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 19:31:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3021df97812b40c091fd46930d43e145\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa402758ace34df1bf535f71dd9517a9 : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b7532c0c3774423abc9e9dd1130847a : SUCCESS in 17m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5f83287977ea4eadb161e8c1af944a32 : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c76ba892ef58f44de9d55dfcb47caa4d59b6b51","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-06-18 20:18:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/419839dc17ca4813b3439dea0e7e0852\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/cf7b4129d4724b0f85c322bf113f1e1e : SUCCESS in 1h 39m 36s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/97a197adc47c4b5a99cddfa91407489a : SUCCESS in 1h 50m 14s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e2be60d58b4e08382955bfd0d26381ba1e1e6bc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 20:21:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f987dc11c1964edaaa22504515620cf3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e407f879d04dd1966b15be5d1a7254 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28bdcadf594a49208ce536ee738b7aac : FAILURE in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8eea9439dd364003b08892845d18c4ab : SUCCESS in 10m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb61885ac6424512a4170a8136d87f55 : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081f9d218d124c499a2e92ac413c8ca2 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a56e23ee2ef34a8bac82fd74d06cf283 : SUCCESS in 47m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01883d7670184fa29cfddd5aad2c5817 : SUCCESS in 1h 03m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/348e2aaacbb54d56a331d8252fd4ac76 : SUCCESS in 1h 26m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3fcc29ae9224130b00fe90a2820d414 : SUCCESS in 37m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dc6f5dbf0a34d44a0324c536938ae11 : SUCCESS in 8m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3904f91ba57443c9b45c33dbeffef45 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bcd992856a7c4645a91690050a7ee9a4 : SUCCESS in 4m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63a2e4d75797407b92f48a5fea6537e5 : SUCCESS in 3m 14s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f642cc3b515747b9a959bbc4bcc7c434 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d05e559ba03a4998999a01480a2df50e : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fbae2c537a3d4cbba91c1cda61dd16b7 : SUCCESS in 5m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a3c643a444814125bcb2c74c2033922e : SUCCESS in 11m 05s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/414d3d3011924ab99120289c5c0de755 : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/db282dd6a7ec45f299a4744c9a7a5617 : SUCCESS in 5m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/333f540a430e47a4a5acf20907bdb5ee : SUCCESS in 1h 49m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb5295147e7942029bad8e32c5fb7d01 : FAILURE in 2h 31m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8dbb09dd04474f97862609314adf541c : SUCCESS in 2h 07m 55s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7fe74c61e75d4e9d80601e12d82aec93 : SUCCESS in 1h 52m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e8344d967cc643479f380659d76e069b : SUCCESS in 26m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c45b3eea70334f67a4f9d2ce680f37df : SUCCESS in 42m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2a5d221050294e78be2e4d1ba7d35eee : SUCCESS in 1h 25m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1b428d1e04df40bbbed8c24a15178af2 : SUCCESS in 32m 31s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e8117d0f9f3348a5975900eb007ea876 : SUCCESS in 55m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c6f687ba5cf47f3851daeaf6f86c7d3 : SUCCESS in 30m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7c4e3ff0353534409a19e31fe057a150e7faaa14","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-18 22:15:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/1f1a4d782bda40568c4c1d594cc1f2d3 : SUCCESS in 40m 56s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993979\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993979","accounts_in_message":[],"_revision_number":1},{"id":"46d8ec7fc2ce03ddf0875232e0ec64b3a9ca5693","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-19 00:39:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/098f8436bd1a40018e588da21cfc37e4 : FAILURE in 56m 30s\n- cinder-storpool-tempest-iscsi-multipath build/bc6e7b14830145a1a9afe0a31ca76a62 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b1a5c3f760982b8f14fe24f718c72693c506fe32","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-19 05:31:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85e88e9d898818aad32a534b7f7a07b24c37c5de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 06:21:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eff4ea3f8d64cf58b427f360aa21b71\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f94402746e044614aba561ff4b832bef : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/851e9aaa5a52431984bf692c7c5737e6 : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1c768309aa7f4a52aaa373e45996402b : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f862baa02722289378ce2f6b897931f6afee72f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-19 07:34:22.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0443d6cce1544dde8627db6d4f8820a4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/602af7f7da904edfafae4196eaaf96d2 : FAILURE in 33m 31s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a893853a29ad44f788ce3ac84515420c : FAILURE in 35m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93e4b7712d8b4837b466a675227d5fcb : FAILURE in 2h 01m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/257981174e83411186f8eefa10ee71d1 : FAILURE in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"297317203482517151a8272bde6223bf139893fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 08:02:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2fbcf82c5244460a320dab0e62ef309\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f07020180146ca9b3c7e39ebd69d8d : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b968ede7a59044c9886777e6f58ff325 : SUCCESS in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4328d7f92da343cf8f81d0515b50e5a9 : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/473823def1c9456ebb273083f39065a3 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cd000910834c7ab2c7b12dcd655675 : SUCCESS in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/e3648fdeae814414a8fd3a4ebaee3bf7 : SUCCESS in 34m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b841687658541c6af663da50ab9a4b6 : SUCCESS in 1h 06m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9df5c87dc8a9463a9f2b0e38afd7a0c4 : TIMED_OUT in 2h 07m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1cedfe88e4c47b68f99f73d8860a413 : SUCCESS in 37m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb20f3888f1545e4900f58b8c8be9d47 : SUCCESS in 6m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6fcea758cbe64362bfdfb5c84d9b69c0 : SUCCESS in 16m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e80fdc6119d148f9af69becb95655235 : SUCCESS in 4m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6f9c92b3df424941a643bcf57f18c320 : SUCCESS in 3m 45s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/0de2d4f820b94fe89800bf027186ef8c : FAILURE in 20m 22s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b1b0c10bb74044809eb62cf46b64aa16 : SUCCESS in 4m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9979c4640813484aa0945f3dd184c003 : SUCCESS in 6m 27s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5b22eec2752e48dd8c6d3d77081e4b9f : SUCCESS in 8m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/262185ebde934d5aaaeed8d52f003c7c : SUCCESS in 41m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77dbc92ba65d4c51ac091a700e91aac2 : SUCCESS in 4m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5c4b3f640a3645cd9097abe5d0a426f2 : SUCCESS in 1h 29m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01878453e513411493e89342f596bbbc : FAILURE in 2h 17m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02f0a5b0b7b44defb45b2c8d62fac55d : SUCCESS in 1h 51m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c6b447377b3e4cf1a1f98802e8cf2734 : SUCCESS in 1h 49m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e327886c75e242a587e0db8ba14612dd : SUCCESS in 39m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ceaca758dda8434e8cc334e498e58180 : SUCCESS in 54m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd6a5245b7b341bfa590c2774c8da4a6 : SUCCESS in 1h 11m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea27ca75216247558c044193852b5a46 : SUCCESS in 21m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c41ce9a4d22458a924749789e199444 : SUCCESS in 1h 40m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ecb2da62994437eb3fdd255d448dd8c : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c8357622aa398666d7791005294a6e7ed4fc3bd1","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-19 08:25:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e400c552cab0e2a875a9bb4624cb114303519449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 09:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4951816c4ca14b3fb5a4c536b5bcdbba\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da66a830141f4f20b17049ff6a660df1 : SUCCESS in 19m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65f5980bf33c40819fc500c7a7563322 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c98095225f8642e1b69f6d5840df38ea : TIMED_OUT in 1h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b88bdf3e652c4754db88f638a3737930aa56643","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-19 10:03:58.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/36a9a4d7f4e44820bb2070396206c04e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52223974d2004d418cde31fe541307ab : FAILURE in 36m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/59ee7103823e4644a31fd7319a8e1176 : FAILURE in 33m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17e909cd3b1e4510a09c2fc17f9ce63c : FAILURE in 1h 38m 07s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7564de713d8420c8323440fbc9d6d19 : FAILURE in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"634465bf56f0e049331a696786a34d5239b4e4bc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-19 10:12:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/c8d20d6d8c7841c9bbc0764ef2af6b1e : FAILURE in 56m 53s\n- cinder-storpool-tempest-iscsi-multipath build/eb1ff60e6f6f4c96a36c03d3398afceb : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"43f4605c5879c35594ac4b0d4fbfb04fa80af86f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-06-19 10:12:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c6eb0abe20462f0fe43acd63b9653784bd6e8103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 10:31:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3f8b763c7914945ac3c3523b256817a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96fc95d944e14b3ca2961555e0691972 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21a0729be05646e595f2360dda2492ae : SUCCESS in 12m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e530923c2c344294a71b8cd987a4e603 : SUCCESS in 9m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2da0cdab940f49d087c958ee984715b2 : TIMED_OUT in 40m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2a6b3bd6d34f4c9faa441193d8cd63 : SUCCESS in 14m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/68895e37c3e84b0297582a6e8b26cdba : SUCCESS in 1h 06m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3af701f487204ddd84ca1cfa2257ca79 : SUCCESS in 35m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0855ccefec7e49fca60b1690ad67a5d3 : SUCCESS in 1h 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/859e71a86c2c4f8aa4a6513ebe875cd6 : SUCCESS in 1h 06m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf7473e7061a47d9ad5a98191fbcc801 : SUCCESS in 8m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7251353a54445a097611170253e6672 : SUCCESS in 21m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/88c28dcdcacd4d69a53b46963aa8ddc6 : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11877672711243f7a2a6c099467b79ac : SUCCESS in 3m 10s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/d8a0113a747b4fbfa0bf4c095a9547a0 : FAILURE in 13m 16s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/232e611b02154bb3a808dec75a6aef5a : SUCCESS in 6m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/833d2f6984e64787a9c65f1da9b161db : SUCCESS in 4m 33s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/81969563d4524fb5a2e287c8cd245a82 : SUCCESS in 6m 39s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2dfc36bf87974baa9aaa860452400944 : SUCCESS in 41m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e70de2862a9440f5b14d57d532e8e8a1 : SUCCESS in 3m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3c6320ff776f4004a9d05e5a704937cb : SUCCESS in 1h 50m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dde5c4915cb2454e8eaeb82218590ead : FAILURE in 1h 33m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8b243f4c67124b72bd40f11ca435ed08 : FAILURE in 2h 00m 18s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e65b0af7305c4c36b04851f143464bd5 : SUCCESS in 1h 53m 55s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7ee0a97e2a924dc29f283cce9dec9e3f : SUCCESS in 39m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4b25bd7ce28540f9a87de4694c1ec6a5 : SUCCESS in 1h 16m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b8549c4900b246a1ba5b35cb6f50ae20 : SUCCESS in 1h 59m 21s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a09df772e944234b23d148608343d5c : SUCCESS in 55m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa6d6a7dac7846469d7667837f90abd4 : SUCCESS in 1h 11m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18aa98dc06164603b09d658096a9ed07 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1b56536fbc5348c069ed40e4b0bcdeecc543c29b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-19 11:08:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fe0befc03b9085aed5ecb3d07cfa7530df9e08c1","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-19 11:08:56.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/250 : FAILURE in 0h 0m 10s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"e1d4756b64f1dd7d557236392ffe216171926c84","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-19 11:11:17.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b33fed6eb1f04d4ee412cdff764f2574d1d4906e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-19 11:18:52.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/518 : FAILURE in 0h 10m 9s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"cd0c0394958716bb52435554b149ff2b3fd7eb34","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-19 11:32:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/deee4a0045da46f782002eff853794d4 : FAILURE in 23m 19s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"9f82f2fbe127fd4040ce4dfde4c2a3b6a2670f8f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-19 11:34:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"149eac7d0e33350a3d2d3a5a74422e67b8008d5d","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-19 11:46:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-fc-driver build/b72a778bfd5b49938c294db889addc39 : SUCCESS in 37m 36s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"ca83634ebb09ab316a4f7db48a6f61e7fa5d9d9a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-19 11:48:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/24d911ca8ba64fc49987e3dcb660df2d : FAILURE in 39m 10s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993979\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"7666273bc06c753d2b4f2f5534be50f7c2217587","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-19 11:49:20.000000000","message":"Patch Set 2:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/164 : SUCCESS in 0h 40m 39s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993979","accounts_in_message":[],"_revision_number":2},{"id":"bc0cc60f6447e71f4559abfe0e8058e0af9980bd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-19 12:10:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":2},{"id":"aff59b740fede824c4fd600f597c3d28b2a56027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 12:28:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7913a0ce07f14a67868434e0f4010b55\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3f8c14defbe4b7badbd343926a66650 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d21d9cdebecb468a9aee74e843f50ebd : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e1823baad30442878e5f01e1437e25ba : TIMED_OUT in 1h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c617e3e8e0a14bb3b7829f22413b0d7a8b7ab4f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-19 12:39:50.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b26c0856e54f4b15a85a9543658c9a3a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6bbf00d495264f6c98c5de23994cf1ea : FAILURE in 36m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0f3701db6c64b82b42b3f6421843d1e : FAILURE in 40m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3c2c151d2a4433ab795b30e8c5200c7 : FAILURE in 1h 30m 55s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/186b02e3bd5a47f4aa2c18febc37a6fa : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":2},{"id":"541ccd81edc67b358e4dff6b44b0eb9b06d46283","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-19 13:07:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d5a5d8d57b88436faa6f3ea1fabdd606 : FAILURE in 56m 55s\n- cinder-storpool-tempest-iscsi-multipath build/47496f5895a540f09ef0591ef67b5613 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8afbd8ee6098e44899aaba7d01c404330737f1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 13:16:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0941ba94004aa1a3cc2a14f487b022\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460b44584e5343908b16ca1fc7ec42f8 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6472566f7ac347739c139e06a1b4d231 : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a390fd58d0f4458bd8dabb552ec1d45 : SUCCESS in 11m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e38e8853af948479678e0baf79fcd04 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9af4f183e19450db87b170c2e4004b8 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c0e4472462264e9d85b3c03f7268a773 : SUCCESS in 1h 06m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04db94cf00ed484484478440fdc30162 : SUCCESS in 44m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49aefba761cf4b328bd5a44dcea7018c : SUCCESS in 1h 37m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb883323a89541f0b95f01f2757c9c0d : SUCCESS in 1h 02m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b3c2f2264e4f7eb1836b8fd813d801 : SUCCESS in 7m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/577c47d9b90b4be7b623893f830ebe9f : SUCCESS in 17m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7ce31d40e2294c0fb0884727c5f2c60f : SUCCESS in 3m 11s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/650824511a2640faa1c19cb4d9c20ff8 : SUCCESS in 6m 30s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/1e60396620234e978628da8e4338b67c : FAILURE in 9m 23s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a398104b5aa4d9fb37b01aa12f44747 : SUCCESS in 9m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0311ee99644541419b5eca9dcb8a2964 : SUCCESS in 5m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/fc51106848ce477f8c27fa5c15d804b5 : SUCCESS in 8m 47s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bbfc3225d8c94ad590019f9212e2bcdb : SUCCESS in 42m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1aa85fda63814bdaadb60943950ab30b : SUCCESS in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/59c6322a18a640de8208124f210ac1f1 : SUCCESS in 1h 05m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec2066c32ea14184aeced4c48687781b : SUCCESS in 1h 21m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ece353ceaba74c17918f67391b265e2c : SUCCESS in 2h 02m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0e3ee0aa130d413cb8de78157b4ce1ef : SUCCESS in 1h 57m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7ba7334ca62e4436b3c573e6a45fe5c1 : SUCCESS in 23m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f65a7c9dabf045f9b77fad7371673102 : SUCCESS in 1h 11m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f8bfc6094fda422e8f380b764e11a9b4 : SUCCESS in 1h 24m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/332b3862dd344e9cb191f2f9594f36b8 : SUCCESS in 26m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17a87149499b4caba18392d02e7890ee : SUCCESS in 1h 54m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d9306d330e5481a8deeec14c1ec7b36 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1664908964e1c2dc2221def4dcc0741b8a45f75d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-19 13:27:40.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/19_06_2026_13_20_07\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"b185a49c8aab4496b1c9303f2662f69ca5bbb136","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-06-19 14:01:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/47f1cc343fc44751a11402a0947956f5\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/239c6e76f89d40b2a4c4b50d4f8d5d00 : SUCCESS in 1h 34m 36s\n- powerflex-v4-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fbee5fded3894757abba871dba39f910 : SUCCESS in 2h 26m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f6621bc796088dca434cb1918c9773761fb63ec9","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-06-19 18:50:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1da8eba210578534b2784f3944759ff4be635823","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-19 20:23:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-1-fc : SUCCESS in 1h 4m 22s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1b27d354fd1a794d0980f7f01bfd47ffec4d3227","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-06-20 03:23:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-2-fc : SUCCESS in 1h 51s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"104bf139ad49a4caa3f2df241ef6cb36abb881a9","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-22 06:56:44.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"52d27546dd99ba0e4b790995f726879f3779c315","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-06-28 14:09:48.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7667f0c40f3bf4ec1b17d47044dbb2d2bc117228","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-06-29 10:48:52.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8ec45c76791cce44d01591b0c802b4ec398fe6ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-25 09:46:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e7d715b99c704f091dc1ff644f9acdbffabb8746","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-08-25 09:47:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d9e7522cf3e797f65f779a023605fbf1b833b477","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-25 09:56:43.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5248/.","accounts_in_message":[],"_revision_number":3},{"id":"53e30664f19b69d98b9e8b2f51dd065fd0388bef","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-25 10:17:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/09a2265829f54685bc3e696e2f39adbe\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/a68be99497ac4d2abdd441c9d223e146 : SUCCESS in 29m 29s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"678f31f50012c4634e9f454a92dc03f0bd38a144","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 10:21:25.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2ceb21c280984ec0a528c20c1a889c6f/","accounts_in_message":[],"_revision_number":3},{"id":"488596d02ebb8cd98f9173fd685fe8087a5e1497","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-25 10:35:02.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/934 : SUCCESS in 0h 47m 47s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/993979","accounts_in_message":[],"_revision_number":3},{"id":"fa5bf8d30b30f6494a9ee729451e7dd86998d0b7","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-25 10:35:19.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1027 : SUCCESS in 0h 48m 3s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/993979","accounts_in_message":[],"_revision_number":3},{"id":"975b387cc05938588c9b0aeb44cfe6dbdce613a4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-25 10:35:47.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1296 : SUCCESS in 0h 48m 33s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/993979","accounts_in_message":[],"_revision_number":3},{"id":"ca2a4594f7a4b019a160f37bf5ccd28f07bea027","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-25 10:37:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2ceb21c280984ec0a528c20c1a889c6f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fab4dfc8ad1a4dc9b77c8df40825acb1 : SUCCESS in 43m 33s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f0efba0d8bba4778be659cc0ad1e2319 : SUCCESS in 34m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/941dbfa8c6624108b21927f029ace35f : SUCCESS in 48m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c7fe997b0ed4ef4946141545ff6b146 : SUCCESS in 34m 08s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/04f40c7ebb58429a80c837ec93e11ba4 : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d029731a46bc2ff5c1b8be772ca8aed6e528a48b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-25 10:41:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/a159d7e91c334a10a29cb5cf42eecd80 : SUCCESS in 49m 28s\n- cinder-storpool-tempest-iscsi-multipath build/e76a8ab58f194b93a5ac67a5428147c0 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":3},{"id":"2d29ab39ca4c3da0a7125ae883e51e376d48ab06","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 10:45:27.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":3},{"id":"02ea6905437f50b2d11361c0adfeb4f5dab8fe0b","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-08-25 10:52:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/0d86714bf7704ed5be4b0f7705009f7c\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/7d9fd1635d714c9f9307c5274dfa2ebf : SUCCESS in 20m 15s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8f81067dc8b34612bef043a7d2ed991a : SUCCESS in 28m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"306fc2a5f7d7902c63459e9facc1cf473078c7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:24:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9663cfc096154c6bb5dbf778a0a4234b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2e79547dad4726b75002ef7082d966 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a88efe1a4904787a591d78fee0c6e90 : SUCCESS in 12m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2869a99bc5e49c1993d2ccda248b3b6 : SUCCESS in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a2c0af7a34349cea582a6c1becd86f3 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb3b7c53608465aa9f812348f3f7cbf : SUCCESS in 15m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/6d29df0768074450854720bec4be4b7b : SUCCESS in 1h 08m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d832750b9f204e9b9941fe1b89b0fe7c : SUCCESS in 1h 17m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76fe1eff14df41eb8d902e17d2a20074 : SUCCESS in 1h 06m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40d410fb2f0a431694e000351340400f : SUCCESS in 1h 02m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c1c88733ee24899b4fb7c895f01e854 : SUCCESS in 12m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/906cd81dd80241c49786f6699893ff4c : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/684f28b4b03f49dd818103a687768ab8 : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5ab20272505749e8ac24133ff3d04447 : SUCCESS in 7m 02s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/dbeb69deaf2b4dadbb99d0674399746c : FAILURE in 12m 59s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e63640357ffb4137aa8f5bd69e72a3ef : FAILURE in 10m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9532cb5dea984b9bb7d44d3dbdaa5ef1 : SUCCESS in 8m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3c92f66020714a1e8372240487599d3b : SUCCESS in 6m 14s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6a1cada363a54894947a4e5b9a3f134a : FAILURE in 5m 45s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8b08c573bee64a44b029f0d85ec2450e : SUCCESS in 8m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d7d9065d980c4b8197f88298f0aedc88 : SUCCESS in 38m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cad4fb063c94ff59d1cf06b6ea7c56d : SUCCESS in 6m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2fdc10f25e784dbfb103b948cf4f6a6e : SUCCESS in 1h 55m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e571168d5ac1474bbc24a02c1eb96568 : FAILURE in 2h 16m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/771683a806a54c659181689bf0ebe9c4 : SUCCESS in 1h 38m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/be99480d749047f3a099b9e026501d7c : FAILURE in 2h 06m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/d3c23c7c9fe74c6aa450e650174f47df : SUCCESS in 1h 54m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/07a5082fbbc5479dbd40cff5eac6a4ad : SUCCESS in 36m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4229e5b2a3244a395a1a0f1973f6c66 : SUCCESS in 1h 20m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c1b87785f2a247d79e1792b3de06a55f : SUCCESS in 1h 21m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fcb5d07e94c3486085eee6f0e74d3999 : SUCCESS in 46m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ccfc014a8f9047c28a1586645da0f8f1 : SUCCESS in 1h 42m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14b80da05f4743f395be8066742f0997 : SUCCESS in 55m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"248b3a60e1c5ac58adb9acf0ffe959a777048f6d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 12:35:18.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_08_2026_12_28_15\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"87ce693189042e5760b91eb5c1af995cee9277af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:47:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1122004161074d06aa680059e43319d0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/371264dbecb44dd39631a983145d06bb : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/de3de09a503148bdb9db7e13d47c0155 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b17bf52aa5a545e4af4809aee7c375c5 : TIMED_OUT in 1h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2acd1720aaa4246f85d9fbe478ba77e4a99a3c6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 05:22:38.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5072/.","accounts_in_message":[],"_revision_number":3},{"id":"e283e9be1e1ff87806dbefa49343b428360ccb1b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-08-28 11:36:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df2909d56ca7b5579936a3b1998888c9fcb634b4","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-31 03:45:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"05cfbe2169311d4a3cbef029d3416f4f531d96de","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-01 18:38:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7995b55a2fc034cd700a4fd78877fc89be0dc4e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-02 13:30:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"91a3751234fb9101a15b6269c657d4c29d30243c","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-02 13:30:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be91caa420d0114dac7b8d1a965807d346b40391","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 14:11:08.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b219763a38964b3b9cc051769c2343fe/","accounts_in_message":[],"_revision_number":4},{"id":"d30690fa25b852a74aa33205493e1b18bb2c13ee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 14:23:58.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":4},{"id":"608efdfda9465316c41c2383e869096ca2ce3f56","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 15:01:53.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1146 : SUCCESS in 0h 44m 52s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":4},{"id":"79d628e4f4332e3e1cfd518527ee96f8324b0e24","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 15:09:48.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1414 : SUCCESS in 0h 44m 46s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":4},{"id":"fc95e239f59939eff92bc4cb2b003591994ddc4f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-02 15:12:42.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1053 : SUCCESS in 0h 47m 4s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":4},{"id":"ccdcab6dbfd2bf29789da4bbf0890793095e9fea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:21:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/756edee94d2942e9a5835811b6270361\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b71300f1af8481fafff8733e15c2aec : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/33003a68a5294e8da20ef578110fe221 : SUCCESS in 14m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/773b310997a54239935c02ab849a29ff : TIMED_OUT in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e38afdcc8af96d1361c75307c507ac4fb9ddf985","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-02 15:26:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/9d881015354a4a92a6d9f9f148663cb7\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/cdcd6e7040fc46e1a6c0fc477cfe1257 : SUCCESS in 1h 55m 00s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"69eafa7d614edf1172092b0927728561738e3329","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-02 15:28:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/b1c9cc81500647b4a97a9488e01f4c28 : RETRY_LIMIT in 9m 39s\n- cinder-storpool-tempest-iscsi-multipath build/221d2ef1d10744a7b813ed347296f50e : RETRY_LIMIT in 10m 23s","accounts_in_message":[],"_revision_number":4},{"id":"1cfadcaaa04205128a622545b9df11cc58e3621b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-02 15:30:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b219763a38964b3b9cc051769c2343fe\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ac8f7ec5f1804a04956f6400e700fcf5 : SUCCESS in 39m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5544ce547cd4acc80a22520a91f5d02 : SUCCESS in 35m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/031a9391cb154a939b5547a9080d6fdb : SUCCESS in 46m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/593bf0fd3fa5441fa046ad4ea0774b0d : SUCCESS in 42m 23s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6c5530925aaf4510aae00783b750597f : FAILURE in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9dacba07622e4195d8e01cf90fb10cc0b2a4527","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-02 16:12:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/9f032b1355514e3aaebe829b2d4da641\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/11e560623cee45db8896ff7897cb2068 : SUCCESS in 2h 09m 46s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9311792249234ba982b510cd7d5a0b57 : FAILURE in 2h 41m 39s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ac1e2edced61eec2f44e9e8444a639e087a76592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 16:19:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a791f9f75d4c7fb3b418a24d4527f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3119592cc51649f39ce177f8522d694a : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b6dfdfefabd48e8adca2fb3e82facb3 : SUCCESS in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07ffc38e1c084cd78d2c34978767b2dd : SUCCESS in 11m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7cdf042a77d436caeb332ecab55887d : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c1a3bb966049398b02a1a2a71b545d : SUCCESS in 15m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/b49b6e51a9c04099937f1f7d2bbe5bb8 : SUCCESS in 1h 04m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94a1c781ed4640dbbcf589277162a851 : SUCCESS in 38m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d99f863a6cc4d02857461441067c6b6 : SUCCESS in 1h 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e910d12bbc37400e93839d811f281b12 : SUCCESS in 36m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b4e4bfc8a1e4c5bbf7ff16603e3742c : SUCCESS in 10m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4555046dc63449b79e2aaf488189fe99 : SUCCESS in 18m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/88664364395442caad945ec7bbef8492 : SUCCESS in 4m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e84a46a14f214ab18c0dc5bccf395a29 : SUCCESS in 8m 32s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/aa5a108a66db4c8ba1ee9aa5441032ee : FAILURE in 8m 09s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/202cecb2127f4f4c9e0815b2aebb5794 : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93481f87263844af8ae7aac0987addd0 : SUCCESS in 5m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e7a5800cb4a64cb3923171706d94c13e : SUCCESS in 6m 29s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3ee3f97c195f4fea8453032c83af5296 : FAILURE in 6m 25s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0f4976deedd94e408c03617ad13ee09b : SUCCESS in 7m 46s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/130def42d0ab48e38174f9f8a4bccf88 : SUCCESS in 28m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a492de252424eda9b2335821093a3fc : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7fc7af1387643f9b4686e4435c1f561 : SUCCESS in 1h 01m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e5bed27b1d924d34b61283e3bb0f3f95 : FAILURE in 1h 16m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/62864dff2c184b21887f0e66a89bb069 : SUCCESS in 1h 37m 45s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3a102ec34d714f77921774712ec18417 : FAILURE in 2h 23m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/10917467974f4fd289aa56fd7b309b8b : SUCCESS in 1h 45m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/600a440dac1944769490b1776369a571 : SUCCESS in 27m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58ddcc96904543038b5eaeabbd22af5f : SUCCESS in 1h 01m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a023236249094ea38e5c2cd625a46018 : SUCCESS in 1h 25m 13s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9c1cf87f568d426d8fbe9b2c237aae20 : SUCCESS in 31m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab73be51520c412fa27e9bc7c2d11938 : SUCCESS in 1h 15m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab52dafd5481479c8ab63cd651a02c45 : SUCCESS in 50m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"6f3ac113df7706dc2c8b36034f8ba7507bea73b3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 18:41:08.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5379/.","accounts_in_message":[],"_revision_number":4},{"id":"5b776230ee6ade7252cda4f3969753013d47ff6f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 08:09:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-4-fc : SUCCESS in 49m 51s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-4-iscsi : SUCCESS in 1h 8m 2s","accounts_in_message":[],"_revision_number":4},{"id":"25563a31feac2a60b48c6c26c8786c55a354f80c","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-03 09:31:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7d71b60f4aac742ce23ee4cc131c4a616750b62","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-03 09:32:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d4b948f5acf5b257a4c941e7924203b3f0819a3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-03 11:49:49.000000000","message":"Topic dell-powerflex-improved-vtree removed","accounts_in_message":[],"_revision_number":4},{"id":"55f7b5cf1033b5e47ca8e780c91011396b19ebba","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-03 11:51:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/1f67851f19af46f3a523e13943d26172\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2e42b1a6d9fc459b9b210f22c40bb7a7 : SUCCESS in 1h 40m 32s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/4e8cd9e50a2645b29ddf32559f0ae840 : SUCCESS in 2h 18m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"0afb490a663f2c00e898af1b118b2ca47fb98a72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-03 11:54:25.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"208e92d011fb9fc1039b353a1dd5c03105603f34","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 12:35:16.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/59b727f343894bdfa77e9f34b8a17190/","accounts_in_message":[],"_revision_number":5},{"id":"2fb7c1beee51f2db4f3353e021eaa02d1eb1095d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 12:47:48.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":5},{"id":"fdc810546f909950a800ccaad27f14bf3531cdca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 13:12:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c8ee3cc660f4b2c9f8b460ad4ebedcd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c8a1b6b59f8449f80383980922f6467 : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/09f2599b74ea448b95c682e96d5db40e : SUCCESS in 20m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1009ee71e7324b638323c6a07a405c53 : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8ef9e08ddfc19bfb652ecc1adc9b7d3bc3991445","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 13:24:09.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1068 : SUCCESS in 0h 45m 30s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":5},{"id":"cc629efd84e9780bd4c039722e68dba8741b06f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 13:24:59.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1161 : SUCCESS in 0h 45m 22s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":5},{"id":"c16920a61584ed8b02c55888f0308db3ad02f992","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 13:28:28.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1429 : SUCCESS in 0h 47m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":5},{"id":"60c0258d4f4d2d1ef3771dfb077ba57cbb2d2016","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-03 14:19:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/fb15506acd6a44b79ed287cfd37dc7c9\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/1915f096fd3e4195b30c5000c118ab18 : SUCCESS in 2h 23m 24s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"bbf1ce939de02cf9abe006a819e55cc493c314c6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 14:38:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/c8e230b887e543818c7fbec60e260028 : SUCCESS in 46m 41s\n- cinder-storpool-tempest-iscsi-multipath build/c41e05cb5f3e4be793c81b3c910832dd : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":5},{"id":"33629d3be6c2799967c57ded3fc30581a7bd2593","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 15:22:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/59b727f343894bdfa77e9f34b8a17190\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/965aa3668e05451e9836443e28bd67c6 : SUCCESS in 46m 12s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/098202ac8d0745b2a42662b16b6b97b8 : SUCCESS in 41m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9d7d03af45a947b7ae5e60d4831c28f3 : SUCCESS in 47m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eeb1da2ea5824ea2987d6a555048d9fe : SUCCESS in 40m 23s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c711174d867940ab829ee534532909bc : FAILURE in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8108c2c05b97aa187af10814d6f47e0974d66bbe","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-03 15:40:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/efed1225b3f34113a90ef1f03eb05afd\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/53b7298f8fe5434aa68895b50bfa5f57 : SUCCESS in 2h 16m 06s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/5611e5728c8e4fbd9582c980ea65136e : SUCCESS in 3h 45m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"33a1fba2655f71973aa7ca51c80576a227e2f3b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 15:58:44.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5396/.","accounts_in_message":[],"_revision_number":5},{"id":"44f63b847ac48db2de8538c2f6d256571a9a3513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:11:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bec5e495201c409a827fb7a3da6dc5f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ef9574b08c44e4855779ec9c905b76 : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc4a1bd4cedd42c6aa67def86b74cf4e : SUCCESS in 12m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/18c2e2d8931d4a17bb2733c1f271c37f : SUCCESS in 7m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d0cd45075944520aa5df816245ee605 : TIMED_OUT in 41m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa0d6cd07dc481ca1077d6be194abb6 : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/58cb7807e36643a3a5f68845dbf55034 : SUCCESS in 49m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df7ad730a7204ba8b13f72eeab4eec5b : SUCCESS in 38m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be7127b4c4d5435db705cca8390f1963 : SUCCESS in 1h 17m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4941d41a8e2044e4a8f90e857a06a2e5 : SUCCESS in 1h 01m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/102708310d1446abad1b454e8d84c8b9 : SUCCESS in 10m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4265fbdfcbaa425e8d94a1a7455a52e1 : TIMED_OUT in 41m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6409a1e4f52544dbb9fa86747ab6185f : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/62b2a1306fcd4ca19b1425b0ae7da88f : SUCCESS in 8m 26s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ccfcd33d4807403493decbeaf19db9ce : SUCCESS in 11m 32s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9d3c8ef0bdc64421b0102cc344a05e74 : FAILURE in 11m 05s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/117cf7c2b0fa4b67be5d503dc1659cec : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7250a86d71cd4b39811679d02559b8f5 : SUCCESS in 7m 32s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/120d4fe909404a07acb94cac3bbdd3b4 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a3ca2a691d524acf8b83951f13473a7f : SUCCESS in 9m 27s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e21ffa718efe43d8901da20c2b31b231 : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a6e693931204f88badccc6536978d14 : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/87d25cc6759d415a848616baf78ae682 : SUCCESS in 56m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d5f40401fc440debf5c0bb1b2be7f91 : FAILURE in 1h 54m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c5b409833524684b52c66cf8e3fd964 : SUCCESS in 1h 51m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c63ff12bd8024db0bfcbb2243ef06855 : FAILURE in 2h 09m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fba5e2533b5e439cb045872fb66694ea : SUCCESS in 2h 17m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/b1cda28167e24ac787b2145fb665fd3f : SUCCESS in 38m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e9923cf342d046fa8d3581dfb313215b : FAILURE in 44m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1c2250a3a01342c79948033f1f97e836 : SUCCESS in 1h 48m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/88e738242ed44a4d97afe33e65f9a6ed : SUCCESS in 49m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb782f08cd30444bb9609d3a1e80693c : SUCCESS in 1h 04m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/911e530939c24797ac77a7c01f970aae : SUCCESS in 41m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"7a12297d1518258d3bab82163366265a3b532f7e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-03 16:37:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9f427664828e7c82ad105d5f39827030ff8ce26d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:00:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e90b6283c44090b6008393179cf8ba\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/032b2ebc7186464b855c01388f60d208 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b7f1818685f64a5889eda876d94eab79 : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/200cfad76366401690daedfbd44256f4 : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9049977d24c53b06aa8e65e589465a730ad03e98","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 17:02:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/3af322be2ab143b290f73c1499eb807b : RETRY_LIMIT in 23m 13s\n- cinder-storpool-tempest-iscsi-multipath build/bd80f7fb760940c3874c8d4b26ea9824 : RETRY_LIMIT in 23m 20s","accounts_in_message":[],"_revision_number":5},{"id":"d6492769692aee0497e7d904f8e03d11469a6453","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-03 19:05:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-5-fc : FAILURE in 12m 46s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-5-iscsi : SUCCESS in 1h 36m 59s","accounts_in_message":[],"_revision_number":5},{"id":"460f264fc06133c84c2496d6948962f4982fec10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 19:07:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c67d64a1a9a4d48bde1f04e0974b89e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf56c3681d4a41509cdd6f258a01e602 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8cabd197cad442898e8d927d85549df : SUCCESS in 12m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f7874630785482e913066fa0a9ce9c2 : SUCCESS in 12m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a6f80fdd9cf4b1f90c13b86251185d9 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93faf22a0cf94b86a3b9dadd95ca4351 : SUCCESS in 18m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/b766c71c792443aa8f336e1477e3fe0b : SUCCESS in 35m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ac92a7b5abc4deab60f694f326f92fa : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/663d83b75437437a8490bd6320ac4c35 : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39861710b4c64b61b01c2300d35988bb : SUCCESS in 1h 04m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72704fb705fc47afbf8f45fa739de753 : SUCCESS in 10m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34b0fa833d7948d490a867a6048506ee : FAILURE in 25m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab5763e61d574fca8e5ab5dd786ef8d8 : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/940e2a9ff4e941e981e0227e6ed8b485 : SUCCESS in 6m 36s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c441c787132b464eb5ef3eab94b05fda : SUCCESS in 10m 28s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dbb96e0e60c74758820ffc94550f4bf4 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/98f0f987bd2742fc89a42217378363d3 : SUCCESS in 10m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dde02c6c46884de581760b3cd69912d7 : SUCCESS in 6m 50s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/967039f5d4034c3fba425e5e19889cb6 : FAILURE in 4m 43s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/28c28c069f5543a1b1deff20e0f71619 : SUCCESS in 10m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3cdf4d1aa694406db4142c9726405f32 : SUCCESS in 42m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7356d4bd55d4ab1828611489cc63c21 : FAILURE in 7m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d84cfcb2f0547219020c9087fa5701e : SUCCESS in 1h 41m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1bcf1360bbe043b9a9d50d3dc6225898 : FAILURE in 2h 23m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2c51965b62154a0e8526ac32befad9e2 : SUCCESS in 1h 45m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e00e18feffee47048da4576e49b981b8 : FAILURE in 2h 06m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e7853465a4de4248b22007106dea53ae : SUCCESS in 1h 47m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/93d46b73d15a4b288b3db58b42e7f11b : SUCCESS in 37m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a7ce59aab6754b338b1ac3ecd4a2bb5f : SUCCESS in 57m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8dfb7f812d7244a8a7590bfd01ad4406 : SUCCESS in 1h 43m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86dd91830a3b4c44b9e06c4ae8527299 : SUCCESS in 51m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f26a49e40b0f4868bd34b198947f9a67 : SUCCESS in 1h 37m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2486f0dfd35480298086c343ec018ba : SUCCESS in 1h 05m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"1f7f4c4b560652dbc56a892e10dafd0d1b83f5f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-03 19:17:19.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_09_2026_19_11_30\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"6531b5fd7bdb3f9f90abdaddc4e09bb07da51b46","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-04 04:26:57.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0eb1cdbca935b21c547dffc85619796c7ed1c3a7","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-04 08:02:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-5-fc : FAILURE in 54m 57s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-5-iscsi : FAILURE in 1h 7m 18s","accounts_in_message":[],"_revision_number":5},{"id":"2a215743cc96555516459dcdc4f3f3ba79483cf1","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-04 12:17:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc7136fecffe4093f5f71da107e83e7af8e861a1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-04 13:58:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b209cef62e9f21bf84642cb4f657b580e39bbea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-04 16:29:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5c3258e084ffbf470a51cc967f6723de9a6b3f1b","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-04 16:30:09.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8bd8d52d6fcb817e129878f1534a099b02275e65","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-04 16:30:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"87c3fe9f1d915a67abc00c60454ba997d4065043","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 16:44:09.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/82a40651d5a44b7ea907ecc729187072/","accounts_in_message":[],"_revision_number":6},{"id":"1e824c90af5e3887c77e09df48f6feb99904e74e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-04 16:51:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8bb5a88b2d5f424084c6ebce17bc7fac : RETRY_LIMIT in 19m 00s\n- cinder-storpool-tempest-iscsi-multipath build/103a54bbfa174d04b3fdae92d41b2fbf : RETRY_LIMIT in 20m 28s","accounts_in_message":[],"_revision_number":6},{"id":"ebd0fb201f0f8cf4d7c406ab32548a47c14efaf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 16:57:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f7654fa4a84533b3e29440199c2f83\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01f82fa555614a33bee1145cdef77689 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b04931545ad44cd9b7143eaaa687d970 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ef5de6136ef84270bf3fcadc94771a4f : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5142948f71f14f1030bcbc04950186377b1ca661","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-04 17:21:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/82a40651d5a44b7ea907ecc729187072\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/315c35d5421f48a0b8ddf7aad6548d13 : SUCCESS in 37m 57s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8883bbfc27a5452ca36a2ac4e7033392 : SUCCESS in 31m 44s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/070d0f5b013b40258bea01dee55f2fc6 : SUCCESS in 50m 46s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1f7c8191295044818fa566db9c75cd9c : SUCCESS in 31m 20s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb833c83a8d04c17ad5e286e7b447c74 : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fa11932046c83e17b43f794bc24be651260d167e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-04 17:23:19.000000000","message":"Patch Set 6:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 48m 40s\n- FC       : SUCCESS in 0h 47m 13s\n- NVMe/TCP : SUCCESS in 0h 48m 14s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":6},{"id":"cbd22cf6a0ad563e865ac02c483f9cb69b71646d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 17:23:32.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":6},{"id":"178bfbdfa94ead34525ef63bca3b0a20bf0e2440","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 18:17:55.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5202/.","accounts_in_message":[],"_revision_number":4},{"id":"e471f10164dc7b311e0d30b1102b8eea2cf185d8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 18:19:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5415/.","accounts_in_message":[],"_revision_number":6},{"id":"ff80e8b20256c9daa07284970f01c53d4c17f480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 18:48:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6bfeb3973054b11973072b2b695fb34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3173af3cc0b84540ba1e2669a204068c : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0068eaaa7ba743cca1cb09b93facf7cb : SUCCESS in 9m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1686c2dca4aa4ab28196cf4eb532aeb1 : SUCCESS in 10m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a1bc0436d8b4494bbb9ee691834ef4f : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3a541a3049411fa54e653cf8fc6e93 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/84998175fe064e9b9205e7f329295ea7 : SUCCESS in 48m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11b6554579cb49529e9d597e140bd1e4 : SUCCESS in 33m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f152b708b864b7080ff8f57e0984bf3 : SUCCESS in 1h 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aedeb3ac098b42bfb97380561f8d3f91 : SUCCESS in 1h 02m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13b65efad1014ac291f3a9ac646cc5a3 : SUCCESS in 9m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/66af842eb80e4f40b44fa6c7f44ab98a : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e77c9015480e422c93466db992ad1fb4 : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/778b0c4af075464da53dd0d04d2487f8 : SUCCESS in 6m 43s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/9b5f6a8df0d149f690419c3a00804fd3 : FAILURE in 7m 56s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d5d6f89341b84aab94c4bb01a3fa42fb : FAILURE in 9m 49s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3f46f024d6c4454bf808133743cd4f5 : SUCCESS in 6m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b83297b6d18e4372b24bb5c7b17ece00 : SUCCESS in 5m 57s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e9ad732c0ca9482a87fed9421c70a00c : FAILURE in 6m 35s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/cd2e5fda4b8047a7bc671a774ba4aa4f : SUCCESS in 6m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae329602c6d4485bb469d1d0f05677c2 : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdc89a123e46463597dfd664e07dbdff : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a3114414d320484196fd3fc1214c89c7 : SUCCESS in 1h 05m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a64d7603b77d42ce96af2b20bb968a0a : FAILURE in 1h 22m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/372ff683f16e42e9bddc57b977fb5e8a : FAILURE in 1h 56m 02s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/df20346f9aab44368bea7ea83a36ed8b : FAILURE in 2h 11m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cd73e313833f4faa810c2425289f4d73 : FAILURE in 1h 46m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4dac1743cfc146dea01a13ebc855b55a : SUCCESS in 38m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ac2ef0b585e348368d777e0793192cac : SUCCESS in 38m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a65f2f088b948c398100d23cdfbe91f : SUCCESS in 1h 15m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cd970b2249e241389ff13727305b89e9 : SUCCESS in 46m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f81f0d3e99a644d4bd884d8ec06e3188 : SUCCESS in 1h 49m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae978c59c78c47e4870261bdf0b0a78c : SUCCESS in 55m 36s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"b81c651cfdb5f6452c7dd626d5174aa6f05dd7f5","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-04 18:56:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/11e8929b46904a91bc51a227aa62885c\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/e9909fedbb714062a095340848ab7ddf : SUCCESS in 1h 50m 32s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/2e3602a332884357bdfd98837f6ee341 : SUCCESS in 2h 26m 01s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"c69a7096cdd0134702bb07d1bcc3a933940b5d9e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-04 19:07:20.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/d32cd19657b448f88dff88d61838bb38\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/4c5c5e9e685041fb82b14753ac36de40 : FAILURE in 2h 36m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"628748a199cc17590e345ce5ceceaf4485501216","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-07 07:54:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4dd04cd76860171fefe617f1d997d5a51f9c4f01","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 08:05:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-6-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"3d1d10af002cc507f0244497b7aabddb82fe7dc3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-07 08:13:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-6-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-6-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"c6879f426cfe02398a56a5bb9bddf92e97d39e92","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-07 08:14:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5167dba89c72413cbda32b7b3b4af3a8 : RETRY_LIMIT in 17m 31s\n- cinder-storpool-tempest-iscsi-multipath build/419a29a9dc14438781e913619e3f57f4 : RETRY_LIMIT in 18m 05s","accounts_in_message":[],"_revision_number":6},{"id":"53aa46c9b42bab8873f7d9d2bb6419aca6f0bd31","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 08:18:34.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b70eeafc6b9240f7af9d1a2278c4ef28/","accounts_in_message":[],"_revision_number":6},{"id":"878f1935cb92d564616ee78d3d8dfd12dd29a24f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-07 08:51:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b70eeafc6b9240f7af9d1a2278c4ef28\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/421c1305da724285a28437659d1f22ca : SUCCESS in 44m 58s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/892c80ecd86b4be6a1b324963a0c66d3 : SUCCESS in 34m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b9b0cd8b0ea24cf6be7e15cd6ec77545 : SUCCESS in 56m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b02c27cb5ddd4d158c88d8150f64aa28 : SUCCESS in 39m 48s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3564e4d07b14a719fcedd26326cc28a : FAILURE in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a27abc901ea959ef403869ae77c7d4d2b8367fd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 09:21:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac53df107c74e878e39fbb99f67d990\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/958f1dd85d6f46deb3e460365730ab6a : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10373f5b7439440993244a07c9aab1f5 : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/79ef54307b524bfb97a527fb4a0c8f07 : SUCCESS in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"484d4fdce2b453f0981728f0d84dc4828cf41cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 10:16:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9569e43cf8c3421e94cfcb73f8bbd48b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7150b168c744c59f340add7873a634 : SUCCESS in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6645fb07a3e48ffbdba961e9aed6519 : SUCCESS in 8m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e95faf9e2ad449ecb13796ccb7adfdc1 : SUCCESS in 10m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae3df87f32c441b79927af77ebe52892 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fbb4870b1e45b5b7061bb527d19fc8 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/45e51621dc9c41bdb65237133122ae42 : SUCCESS in 53m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a00e5f30975c4301b392f434b432ea71 : SUCCESS in 1h 14m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/579f604bbc5c478d8d8672ffac0c34c2 : SUCCESS in 1h 37m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01a9c35dbae94f2a897fa2eb47ad2aa9 : SUCCESS in 1h 04m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dc42948e0964397a5bf139f4169e322 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6868d1b0d5024882be4dd77a8e025832 : TIMED_OUT in 41m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0c2d5a960c05471e944cb1ab76dd96df : SUCCESS in 3m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8a576a5cf30243d695e5a398ae70ee35 : SUCCESS in 4m 08s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c6215f16e77e418f999ca51ee230a002 : SUCCESS in 8m 09s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e828831e6ee543be8296e52f06b2299c : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d61640f54d2549ae8f4ba47a8ed93d43 : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59a52491f1174fb2bf420914ea087d51 : SUCCESS in 5m 41s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1876eda3436148a88affc56e8b8b6ef4 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/03871e5b75b842d5944a4ee8ee606c8a : SUCCESS in 9m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d31447f9f0c346b9906dd92129f22023 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81f337d8d2bc4c44881b7768bd9a9ee6 : FAILURE in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/42edda4901cd4cbe955be462b63bd4e6 : SUCCESS in 1h 38m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2af7cba94075439e9c77a1b1862ecd97 : FAILURE in 2h 09m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2da35419479f43509d9bd7254674b66e : SUCCESS in 1h 41m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/0ecbd95cd4324abc98c0dbd5ca4af0e8 : FAILURE in 2h 16m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5c017964cb3a4a09ad16f9529410a386 : SUCCESS in 1h 44m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e1ce5f7647546a196eb63da50afc967 : SUCCESS in 16m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee6e2e4ab26d40eb9ce7d263223cbe86 : SUCCESS in 1h 17m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ae9f91171b740bda9c9b46ee875b55d : SUCCESS in 1h 23m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/194667a254b446e4a2bb4d088e97349c : SUCCESS in 44m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ef09a7a6dc94d5bb018d254937cf721 : SUCCESS in 46m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a25bbc56064b4df59ca015cf1db6667d : SUCCESS in 33m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"58445188177fab24974c1c9fd827b174dd9d5a81","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-07 10:26:40.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_09_2026_10_20_55\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"8abf99558ca5aed8897178e658f4a6d87e52bc65","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-07 19:49:21.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"938e18bc4445a767eaf6f90f6d5353cb5dd5f3de","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-08 04:29:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a37c8926a99070c1963bd4bb69dcc0c43f56933","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-08 15:24:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8ff7547db4a31e96425facf4738520b42935a8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-16 18:34:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"46fdcf60fbaf2d9cb2073e7806def7b98490a148","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-16 18:35:48.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"50e3d657b359bda7140dde21002d393d6ea99a65","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-16 18:37:31.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/8545d42a73d449eba0d6448f265695ad\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/26f46a4f5b404d119289989d5e402314 : NODE_FAILURE Node(set) request 200-0000093848 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"1eb2e6cd5b227173c0d7b1c139ad45d4adc459b2","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-16 18:46:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fc0232adb09c4203a2457aea708b22ac58d96287","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 19:11:22.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/ef1dfee0f97c4aaf9f9440339eaa5b04/","accounts_in_message":[],"_revision_number":7},{"id":"7c42ee06a5c02495191a5d0d8f6fba033505d425","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-16 19:29:11.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":7},{"id":"5d0eb51478060b507a329b6f9445f11e6c00a3a8","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-16 20:25:17.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/08a37215c6434516b6d6fa9d4a70c819\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/474c3239f89442038b88f4c228281363 : POST_FAILURE in 1h 38m 07s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"011a05f7caac7f157c91d92b8151e2670bb8f495","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-16 20:25:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/ef1dfee0f97c4aaf9f9440339eaa5b04\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/815ffce551f74fdb8e6a5b566dab49d9 : SUCCESS in 40m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/94c85bf604ce43aabcbb2fca150f7d13 : SUCCESS in 35m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/328074a2a7074d65bd7a0b0f2ea3e194 : SUCCESS in 54m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a2224f169974d25b2350cc676506f18 : SUCCESS in 37m 02s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a00e764269f34417a6490936a8a46ccb : FAILURE in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"afc8b72cb4c399bc4436df54baca683cfc39a9e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-16 20:45:23.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0b97ef2360464988873d04a246526c7a : SUCCESS in 45m 25s\n- cinder-storpool-tempest-iscsi-multipath build/d57391d2eeac47c9b782743a9b43bdaf : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":7},{"id":"39bc8fe3a619b50a7fb3fabd3b3f4c3d0888dd94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:48:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2749fd80805443f78abd64b9c8c7852b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f04be170a94f79ae716c2fa5dbf935 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e357a59d11e04a15a4f0398fe2403283 : FAILURE in 9m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/470470d366374506b3318df9cc06ea2f : FAILURE in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec19a03906b5461b875f2984b411fcbd : FAILURE in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1939cb58aff14fea8bfbc57965836aeb : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/8a45e293629c48dd9860d0e34ed40087 : SUCCESS in 29m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e7f6c55660a4140a1ce3a92b47f91cb : SUCCESS in 40m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca8400d27dc549dfadc880cdcef3ca73 : SUCCESS in 1h 23m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37a279d878024be58e766bb70b956990 : SUCCESS in 28m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e889ad4f9e7414ebf506ce2b5db7104 : SUCCESS in 9m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e8476ce591cb4eee911e0582c66b0dfe : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bdb3148ad06f46febc22730d7e27c9db : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dbd0391708d4630b9461d926364c6b2 : SUCCESS in 6m 08s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/fd0a3bd75d7c4d3facf1c8f17d6c59e2 : FAILURE in 7m 23s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/945f170b27444ef7a41b977ef1c53a0d : FAILURE in 9m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3ee189322a8a460b924d26f2c4b7b42d : SUCCESS in 7m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9339bef63ccf4db3a91dcbe7d892eced : SUCCESS in 5m 55s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f89a2c6836bb4ea0afdf98aa4f444b97 : FAILURE in 4m 08s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/ec2c0937065745709cb66a3a6490434e : SUCCESS in 8m 15s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/825d9980bcb44120bd196cdb5d4f7f0a : SUCCESS in 23m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5e3e95792234f339d62a1fd7fa51899 : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a424da50d74a41b489fc74157d18c980 : SUCCESS in 44m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8b57bad44b584bb0935c8768344798cf : SUCCESS in 1h 23m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd5e1ca00f044dd58cc195131875337e : SUCCESS in 1h 46m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/c99cfc9097da4f578c535dbbd046ea21 : FAILURE in 2h 07m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/fd7efd5cceec434a9249b583ed419a3b : SUCCESS in 1h 56m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f341ab8969034ac586d2a3d02d692620 : SUCCESS in 23m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cdb30feffb504876b20106e989212d8e : SUCCESS in 49m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f973c45acb614eb29620b2b01d3615dc : SUCCESS in 1h 04m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e20a7c64114e47119f5ec078d3c57f94 : SUCCESS in 40m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50fb677569a14beaa4658a7794d996a2 : SUCCESS in 1h 48m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df4fed56e35d4e1dae99e4dd9d8cab20 : SUCCESS in 54m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"b7c8b11a7e6a3e782f0bdec2cfea2d9dfcca86cf","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-16 20:59:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/c7e93c67f1054e50a1f53ec05c08d1c9\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/1d4f80b9dfa0472fa8872142b0cc6145 : SUCCESS in 1h 48m 36s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/4e81f0134ea54c4698a466aba5af1713 : SUCCESS in 2h 23m 21s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b4b322fc5f58b2ef6f621cb26672c6ff54de47c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:50:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e83a0ec3cf0442da979525c7ad50bed\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/54b0517ea2124681ac957faa6c525876 : FAILURE in 18m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3b5202759acf4027874c0bec8245cd0a : FAILURE in 12m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/802d202048bb4f99b7b90a77a879d5b3 : FAILURE in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"43ad2e32daafd683db47042ed4543c1949581940","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 23:10:00.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5396/.","accounts_in_message":[],"_revision_number":7},{"id":"31f09a62cd5e2d93bf5cfdba97c9bd4255cef61e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-16 23:16:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5573/.","accounts_in_message":[],"_revision_number":7},{"id":"5376c5746509e6f526a570d2bdae885f91e5ec00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-17 05:42:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"8a1e8844981476df66de5f2e229cb1b66b975a2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 06:02:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5ec883a5a354e50a0cbce0daba5c875\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f655e93bbc4467f828cc7ea00835280 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/015bf7d5b8a742bab7e03a8d6a5b039a : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6167efd7cfcf4d94a6e45afa558032ed : SUCCESS in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9c896aa551d8264ae7d64dc8c6d3b1c01a3c9857","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 06:02:53.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1f3dc69dca2740be8c0842e30e2dc9e9/","accounts_in_message":[],"_revision_number":8},{"id":"c2ff129dabc902ea183f753284dac2c57c63b082","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 06:34:45.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1f3dc69dca2740be8c0842e30e2dc9e9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9710556a58744020914e79003c1c0a42 : SUCCESS in 44m 53s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a9f0e0c8973a4d27a21abb7a75d37853 : SUCCESS in 32m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7ef409b9a6347d09bb13e668db6d629 : SUCCESS in 51m 36s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7b7935c8f5c4560af01b7b117b2a6a5 : SUCCESS in 43m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24da703ab1074dad89b472e1b3424088 : FAILURE in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a10f2e0e9700254839d99ec6da82fe473dc00362","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-17 06:36:40.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":8},{"id":"3111197273f8c44489d25c28fe25eb55d3813146","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-17 06:37:06.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0ba97f1ad9e64905a41737a63d6552c2 : SUCCESS in 49m 07s\n- cinder-storpool-tempest-iscsi-multipath build/8a2d264e84db42eca09dc3e5578fe8f3 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":8},{"id":"dce9899d927233388ed29df80a2e2198c42f9562","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 06:53:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5576/.","accounts_in_message":[],"_revision_number":8},{"id":"43dab7b8dc940d8e1d7bc4f2f0ae38ce2a0c17dc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-17 06:56:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5399/.","accounts_in_message":[],"_revision_number":8},{"id":"5152e69ca961963a936aa6d917ce4b6df397a317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 07:34:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4762f457adc464e93da0fabfd271a43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1382a58524dd46ca872d7b1a5a7365d8 : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfecee222d924eeca616ab3b93001d34 : SUCCESS in 10m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b445eadd4f349dda4f498e0e1185a9d : SUCCESS in 10m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb2cb4bd097c4f84a9b5cd42ee4df83c : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea118645b79d4f4fa029c800a6c9be39 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/ca4964f95c0c40fc88b4a9f4e6d35a7d : SUCCESS in 58m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ab0b5177f9d454ea99d96e48a7740df : SUCCESS in 45m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bcbc9b975704d6bb0468cf62648aec8 : SUCCESS in 1h 04m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d230c1e240b84130af3c3b5751245312 : SUCCESS in 39m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e487778dd47439aabe51de130449004 : SUCCESS in 7m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/274ed51ad3b44cb69a7d5668a367a4df : FAILURE in 24m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0744d329e6154119bbb9b21078b9e623 : SUCCESS in 3m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9382110862b543c5aa5073a24106e9c0 : SUCCESS in 7m 32s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c6f7146203264aec92ebd351adefaba4 : SUCCESS in 8m 43s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dd1bbbb681954c03a73ae8aa83ccb0a1 : FAILURE in 9m 47s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c63610ab38124598b58f97374dad3617 : SUCCESS in 6m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e26e476ceff483aa96a06a94009c38e : SUCCESS in 5m 05s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b76ca2245d9f48698f44be40c3dd56fc : FAILURE in 7m 02s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8b55030a45d64059aa02aa347e88afec : SUCCESS in 8m 36s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c3f273690f64d37ab827b1c79b59ff5 : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3282c4c1b4ae41df8ddc015361a43e59 : FAILURE in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e0e2b15da6014eabb2e91c1ff37cfbac : SUCCESS in 1h 39m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ef2d7398437044a4ab7fff70feeb2fc3 : FAILURE in 1h 08m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46f2c2fb38e54343961860cebeaa0877 : SUCCESS in 1h 40m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/0b8f8b798ea04ecfa1f19504a26acdbf : FAILURE in 1h 49m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4bb928582af742a6925530bf7f33d211 : SUCCESS in 1h 46m 22s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/452ee9661b644325a34d2d97d1c74043 : SUCCESS in 40m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e9d830a95134ec4b1befd684d82f1f2 : SUCCESS in 1h 02m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e4e2c6b9b39649f997f58e0faab21740 : SUCCESS in 1h 40m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6b4ddf6819f94a60b7ae24aec77e0627 : SUCCESS in 40m 25s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0b1f68424e649d89bd4d7b8232d96cc : SUCCESS in 1h 31m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/288ed8ef28ed465faff58b6d9c5751af : SUCCESS in 26m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"5ae77df00fefbc0a019a78f26a8538396b33bbee","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-17 07:39:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/e57eb0956b7643298bd10f032517890f\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/f8eb28e03d194c459223519005b79192 : SUCCESS in 1h 55m 33s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"57d8e7f83cc5bf47f7939c688d6d3fd6992051f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-17 07:43:48.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_09_2026_07_38_08\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"06b7b9ee810a8777df767d607abd1e78a868fa60","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 07:45:15.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-8-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-79-993979-8-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"14656ef1fd47f2c8d1eeea2c5c3f03937dd7d19c","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-17 08:05:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/a4c645a8b89c43a7b528f6389ae0a604\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/b160c091ac8c4bb2890d270ce0f78ad4 : FAILURE in 2h 06m 19s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/0ec5c823726240cb97119a8b3be8c704 : SUCCESS in 2h 21m 27s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"faed59746fe6e7eb6afacc7b6e0c40e173246763","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-17 10:49:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-79-993979-8-fc : SUCCESS in 53m 13s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-79-993979-8-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"3a922862bc0692efd9ae3a485051f5a6572af47c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 16:15:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"440b1059c3a6655cf81b82dbb1498be5e82d8df8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 16:15:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b5e75663ec7ec76eb62d703937783114babbbeae","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-17 16:16:09.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9835daf38fd79542e2d9ae9004d880242d244754","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-17 18:12:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6264a0506bef87e3d6bb8531a1e0b58e63111871","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-17 18:25:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"373815b18e4715e0d42398bbcb1f659cfc8f7569","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-17 21:02:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/66a98151b5a34639b3e7ee10f428f493\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/f90a8f599c0f4153bb5fa991e0085aca : SUCCESS in 1h 53m 22s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/521556eb7b69454c9c50d6dd2613e0d0 : SUCCESS in 2h 35m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"d7c4559ec0a52632d6c95a32e4f640883fc1e0f3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-18 14:30:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bf1db4fe8c6d0feafa6b9ed05a61d55a5b7524b4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-18 16:17:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/3093d722c2ad4b7cb573270907f4d78a\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fe9a6f016c06488290f1a14d63b4d0ce : SUCCESS in 1h 45m 57s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"36c65dd16cab97f098a366dc221db2c5d90d84fa","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-22 14:55:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"47770640409ca8da780862ef2c728a2b36b97493","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-23 14:12:14.000000000","message":"Patch Set 8: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"de2cce03763d7107cf5a68df8c79dd7cb6f55b95","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-23 14:15:39.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3388a790a24f60b94136f0750be44c533df908ce","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-23 19:10:40.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0cc0cc074168f8c7d45cb1a57f88ae43ec204077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-24 05:41:00.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"a3bffee16cacd7c37d683f7f6ed99aa26f011f03","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-24 05:41:23.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"64aa5c6c9393de3fc9a3e2aa18c94531b8b4e7db","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-24 05:57:35.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/63db1b26451043eab9e4137f84bea233/","accounts_in_message":[],"_revision_number":9},{"id":"c5b687488f380b5644540f5d82e5deb9acf2fa36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 06:08:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/258036facdfd4a84bad004c2ff912067\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/382d36a6b6b94754914bea8a3794a36e : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb0e39ed633b4dcfb04a0a11c435b7e4 : SUCCESS in 25m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"24fec51b6b001b9ac2bcbf6624262bbd9dc2c88a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-24 06:24:32.000000000","message":"Patch Set 9:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 37m 18s\n- FC       : FAILURE in 0h 38m 46s\n- NVMe/TCP : FAILURE in 0h 36m 5s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":9},{"id":"d9228ff3a049d2ff3ac0198179ad417a98a19eed","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-24 06:36:28.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9945ef96cbb44cd993c3ccfa4c36dc2f : SUCCESS in 45m 17s\n- cinder-storpool-tempest-iscsi-multipath build/12c0df9bcd364fa69ffbfefb79a4db1f : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":9},{"id":"343798eec68d99e1ab9f4e30f44848067caa7f82","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-24 07:41:01.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/2f967ece84f84ef4b0e5f5ec719d3d32\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/fb0dc5888f0a4c8b89cf785ce6bd0a04 : FAILURE in 1h 58m 56s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"5f98a887612e24edcb73eada40935d150ecd2a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 07:42:07.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cdd654205094938be9c1ada7ed3a38f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abb6aff768a94505813d51fd31fef590 : FAILURE in 6m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bb90327cf4f40d496253983796c722b : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84b0863f0b9b4191824b0c96c98ddc8b : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccffeb205584cb184bc5d60a1a3fd39 : SUCCESS in 14m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/dc67f957047045b4874d8dc4ef5be50f : SUCCESS in 1h 06m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae47f68826f74110880613a71f9f54d7 : SUCCESS in 30m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6699689ca7d54a2784ffe720e7e514fc : SUCCESS in 54m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30932fe78d32473c96a93ec67bcf97a4 : SUCCESS in 1h 04m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f03af20ff8ae4f5393c636acb3447099 : SUCCESS in 9m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7536ea8ab6f4630ac6fa3e1269ce6eb : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/083972918fa4431c92b32921363b7593 : SUCCESS in 3m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b03a5de78924e918a960f38f71c25a4 : SUCCESS in 6m 34s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/3f9377cdce744613bb620d2247a3db15 : SUCCESS in 11m 46s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e443e9c877e0436e8f847be37f9923d1 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b566724ae1124e0098654d7124251e63 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/99c312453a6e4fb593d861db6e0f0c8d : SUCCESS in 6m 12s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/965dd7d595634bdc9a33ce2469d81f17 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0df7b5a5f719413fa891b1a5d44c2ec5 : SUCCESS in 9m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a9c38e98191449719f60434ac515b2d8 : SUCCESS in 28m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e431c706211e46bf9011013b8dfe28a7 : FAILURE in 7m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f616bbc2f4184f2eb51c012089bd478a : SUCCESS in 1h 14m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5cdcd27c221d475cad50ae1669502fa7 : FAILURE in 1h 56m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/83ce0d7852864e80ad49c4844432e11a : SUCCESS in 1h 47m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/82c18f9811fa4c01b32bd9f339dbfd4c : FAILURE in 1h 51m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/09d05ec0dfe44956b8d5e380c78437cb : SUCCESS in 1h 39m 33s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8527677cf938402ca460da351c195a93 : SUCCESS in 18m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06f6ea62c9954812a06c31657ca00fe7 : SUCCESS in 41m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f7c9620c312b4de5a8f573a448a23440 : SUCCESS in 1h 30m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8f62e9fd414465e9d2e4c335ad89ec4 : SUCCESS in 32m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a9fd18a94e5c49e39b9830762dd528c4 : SUCCESS in 1h 39m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a05ec75edbf94bdd8cd9faacca12dbf3 : SUCCESS in 1h 00m 17s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"17d714d65966bd3917dc2fc69d4141bedb8c5649","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-24 08:04:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/9838e7464de946e79719eef2cec994c4\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/733ed3de1d6b480bb396d66df2737337 : SUCCESS in 1h 35m 30s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/ecc50276e9c147f49956143dc702d9b9 : SUCCESS in 2h 22m 50s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"00cd5e011ea92e733e3e4306baa7fd2d95ba2e25","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-24 08:32:42.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/63db1b26451043eab9e4137f84bea233\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/149cf4c583324630ae5449aca4971930 : SUCCESS in 36m 07s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c4d541b237d43c8b2acaa1183d11b13 : SUCCESS in 33m 54s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/88a38e794c494d3b826cb378f37e7d72 : SUCCESS in 49m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5ed9472e719f4bceba3e4d4133683648 : SUCCESS in 28m 50s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7d97f4db3334400281e725cad7fe38d1 : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9921e656348022da0f6075811c1fe9985f704544","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-24 10:55:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5889d78c4697a1e67a911450b1995115a7211ba1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-24 12:35:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2cf01b070ff5b5fa35d6bf9eff00d2774a482f01","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-24 12:36:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/aa46efbb4bf9496c9098851c3f3e8f48\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/664b2cc01c0240a3ba30ffe4602fa1ad : SUCCESS in 1h 40m 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"4d6296a6cce4a5fc35f7e17ef41dcb17a9d24cbd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 15:58:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5676/.","accounts_in_message":[],"_revision_number":9},{"id":"c93d3c6118154e39788a4ee48d928d46489c9641","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 17:16:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5499/.","accounts_in_message":[],"_revision_number":9},{"id":"3aa7d040ecbcdd483c827c14fb7011b04423d3a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-25 06:25:12.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"fad2b9d801ff64fd234653a4664665e31c7592c3","author":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"date":"2026-09-25 06:25:27.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a41fcaaba2d305835ea8cd533e1a73784d2f2f61","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-25 06:27:51.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/9526d2c10bba468c9cd840b6cee82ed4\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/870623e8943c43709ccf0f4913ef52ef : NODE_FAILURE Node(set) request 200-0000094746 failed in 0s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"14d0f2fb3040897039b3d3ea9efaeaddd3915af1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-25 06:41:39.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/24da0d476d684ae6a691a19a66c5fc31/","accounts_in_message":[],"_revision_number":10},{"id":"d314987a50a36a7c4c9cdce95b2b7c824866503a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 06:52:53.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d81e70b5a894beaa4e94508cfecfd39\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/525df8caf85b477a984e55a8c61fc765 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f2176851e49d4ec89dceb3c12c1eb73c : SUCCESS in 25m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e6ef6389eb08911bfce100fc7cf116a54882625d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-25 07:09:42.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/24da0d476d684ae6a691a19a66c5fc31\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e63fd6eb71b74dd394ef69f396bbc35d : SUCCESS in 36m 54s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c08a3706ef94a8d8e511143065c17f2 : SUCCESS in 37m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/167c2d11817b45f892ddef9d4aded3f9 : SUCCESS in 43m 58s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a901b95a8a2643f69c15032ce12fccd1 : SUCCESS in 29m 06s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8b8262b5c7e04259b7f18e75c8297aa2 : FAILURE in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"63bf510b0a44f2b5733bd5c8974291ed10a4d2d2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-25 07:11:34.000000000","message":"Patch Set 10:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 37m 16s\n- FC       : FAILURE in 0h 37m 15s\n- NVMe/TCP : FAILURE in 0h 37m 45s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/993979","accounts_in_message":[],"_revision_number":10},{"id":"95c5196f3db3ec826e39540777a46ef197e86ece","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-25 07:38:27.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/3c3761f15abf4115aecc3c91f974efe3 : SUCCESS in 42m 58s\n- cinder-storpool-tempest-iscsi-multipath build/d13d44176bbe44f7b9ba0682e89b79fc : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":10},{"id":"c33bea2eaf69750aee8b2581ed75c4a78eb512aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 08:02:50.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":10},{"id":"8a75adc0d051399eafdad9612c0ca46acd27a5ea","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-25 08:36:26.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/0ee9a7f6881349ff8ec7fd2c2f760470\n\n- powerflex-v4-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/8f830fc9eb8c4cc4bf920ff7c0eb5b88 : SUCCESS in 1h 42m 51s\n- powerflex-nvme-tcp-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/9f7268501621492db4303c1b24a519b1 : FAILURE in 2h 10m 43s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"4790613f3332a1959ffd1c0ef0d3313d71809178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 08:49:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2860982c1504e20b19d569c55ff0912\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0638cd98f6420ba8b3612da31b4bad : SUCCESS in 9m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0969955208246cb8f330b9c0d617eec : SUCCESS in 15m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4add4b2f5344876b9cbb2651cc5f8ad : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1ebb05bc574ca4b0ede5c9b2035479 : SUCCESS in 7m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/1b6221cafcf44df2bab6365c28855cfc : SUCCESS in 29m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3e524cef9314fe899cfbf1ed6d4800f : SUCCESS in 33m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/698f055dbc9e4d06a0aed635595019dc : SUCCESS in 1h 27m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8b7e3ebf5bc4b50b27cbfd0259caa9f : SUCCESS in 1h 04m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59ee962ddc9945dc9e9249f9d32deaff : SUCCESS in 10m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c84a1ecc2679459db9f25b3e7ba022c1 : SUCCESS in 20m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/42e93ab3b0df496caa60a43c9e86b4f8 : SUCCESS in 3m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/188fa60a360e4edfab66d73efc765078 : SUCCESS in 3m 34s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bca52e79c0a141b1ac396503fa9ba8f0 : SUCCESS in 9m 03s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/773994ce9b8b480abf0742fd80c8f727 : FAILURE in 12m 12s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a8316e5bef4544f291b3c01c72a0ba93 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f49e9401ecde4883be1c91ac916c41bb : SUCCESS in 5m 56s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8191e578afc84382a77af8bddee58dd7 : FAILURE in 4m 24s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/99fb0a8ecde4470e97a133e958fdc10e : SUCCESS in 6m 33s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05cd293e617e4c348a35dde429e0d450 : SUCCESS in 39m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bcb4d1ceadad4741ae45b64e74330303 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d73ec70d329e4a23b4c70fdcc81fc67d : SUCCESS in 1h 29m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b1ea09815ea42c3be06fa65f3b18ce0 : FAILURE in 2h 15m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ed6bfe2830b143e59ed3f999011550ce : SUCCESS in 1h 18m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/61dc65ef761a406487f5cf4d9d56954c : FAILURE in 1h 55m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a952200d09e94a48804d7e3ae4cabbff : SUCCESS in 1h 32m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/80178ef6011f43ff830bb9a4fd1fb571 : SUCCESS in 19m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49b3ca860b634fd6bbf5c9b2b27c0f83 : SUCCESS in 1h 17m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3647861e27cc40d2b67c29f313b5cdaa : SUCCESS in 1h 06m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2ac880b7161f46ebb838eef121dc061e : SUCCESS in 47m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f3d84e720d74198b1c114e7663796d1 : SUCCESS in 1h 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/309bc44166154dc48d2efdce04a3e1e4 : SUCCESS in 55m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"f47812f46699199971f822c742cebfa34d457b3c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-25 08:59:06.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/25_09_2026_08_53_08\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"4769f84ae5ac1093b2a76b14937ec58c32c65a07","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 09:04:17.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I77291902e477ed14180dbe4692edde6717bbb791 .","accounts_in_message":[],"_revision_number":9},{"id":"5a6ff2d229d4621a0139b2d58763849f79c4a38d","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-25 11:27:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b0dc1385bc23dbe2240e472849ba0d094c109035","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-25 13:19:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\nhttps://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/buildset/4a19f8d528494f919741035166be3ee7\n\n- powerflex-v5-cinder-tempest https://github.com/dell/dell-openstack-ci-logs/tree/main/t/openstack/build/e2f13e87af334dc79331c682c3c06644 : SUCCESS in 1h 50m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"8744c6d670d37c4a7ccb6ab64acd3492955bef05","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 00:11:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5706/.","accounts_in_message":[],"_revision_number":10},{"id":"81ba513887fd34a4e354cffb8dfb1538f00405b2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 01:09:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5529/.","accounts_in_message":[],"_revision_number":10},{"id":"45d51785c9319738df533da70ccc3e2cb383f922","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-09-29 05:51:23.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6bda16fc2d10135e7fb492c41af3c92107e118da","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-29 14:12:54.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8898583c8891874113727e011af20eb9eeefe308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 14:13:39.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"94dd4b3618f2940c60a06c717ec8f2d7215b6d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 16:07:48.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3839614e045d43cdb6542f1b72afeafd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd18f2a492df4194a0d84ca858ad01ed : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8266261bdbe46dda6d1ea6e0d9a88f5 : SUCCESS in 11m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4e13b906b199436ca03c992d9c9b7cb1 : SUCCESS in 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce4dd6d1b30438db7a8f4e002b34116 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c6bf292cc0a54e64addd59f7c5b4cce8 : SUCCESS in 1h 03m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89bf8161bc4d43a494d6511b5bb9bca3 : SUCCESS in 54m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/726b2a12b885429083a80bbb269640de : SUCCESS in 1h 43m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e2522bf80df4b029192bd18fae841c7 : SUCCESS in 1h 00m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/618d355072ab46c280c2f31a846a7229 : SUCCESS in 8m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd1fc74aa5164777a09525fae6ad8765 : SUCCESS in 57m 45s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f78da42ee5fe4b9cafa559a7d5fdd918 : SUCCESS in 1h 42m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/944926b0d8e546a1b2e94a6d02741ee4 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":10},{"id":"ea9f9eef13c28366d3796be9e2a5089349cc3a0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 16:07:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"0be24957641b3e821f9ea94c38f20977cd09ef2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 16:10:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ba85ef2d364b4792a730353b2bfbf8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec225401d2f49bf8223b51dfd953267 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2f73995ad4544ba8a5bd93dc6abe4f7 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fb5e9608d4a8cd1c2513a4da771e566d9cf17d8b","revisions":{"18605da7b7c05f89ebfdc5e06813e3d523099d70":{"kind":"REWORK","_number":1,"created":"2026-06-18 17:44:18.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/79/993979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/1"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-06-18 17:37:38.000000000","tz":0},"subject":"Dell PowerFlex: Add direct-child count for image cache vTree size","message":"Dell PowerFlex: Add direct-child count for image cache vTree size\n\nAdd support for checking the image cache vTree size by counting only\nthe direct child volumes of the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nImplements: blueprint bp-dell-powerflex-improved-vtree\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18605da7b7c05f89ebfdc5e06813e3d523099d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18605da7b7c05f89ebfdc5e06813e3d523099d70"}]},"branch":"refs/heads/master"},"9fb2d9936cfaf41c249e7302a35f88174a8a1840":{"kind":"REWORK","_number":2,"created":"2026-06-19 11:08:24.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/79/993979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/2"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-06-19 11:08:05.000000000","tz":0},"subject":"Dell PowerFlex: Add direct-child count for image cache vTree size","message":"Dell PowerFlex: Add direct-child count for image cache vTree size\n\nAdd support for checking the image cache vTree size by counting only\nthe direct child volumes of the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nImplements: blueprint bp-dell-powerflex-improved-vtree\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fb2d9936cfaf41c249e7302a35f88174a8a1840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9fb2d9936cfaf41c249e7302a35f88174a8a1840"}]},"branch":"refs/heads/master"},"89e299de2fa5cb9ba601faf53e93b4abb6a24f17":{"kind":"REWORK","_number":3,"created":"2026-08-25 09:46:47.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/3"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-08-25 09:46:30.000000000","tz":0},"subject":"Dell PowerFlex: Add direct-child count for image cache vTree size","message":"Dell PowerFlex: Add direct-child count for image cache vTree size\n\nAdd support for checking the image cache vTree size by counting only\nthe direct child volumes of the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nImplements: blueprint bp-dell-powerflex-improved-vtree\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89e299de2fa5cb9ba601faf53e93b4abb6a24f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/89e299de2fa5cb9ba601faf53e93b4abb6a24f17"}]},"branch":"refs/heads/master"},"6c910f56ec205ee9bd17a14e8787ec42fa1793e4":{"kind":"REWORK","_number":4,"created":"2026-09-02 13:30:00.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/4"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-09-02 13:29:26.000000000","tz":0},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nCloses-Bug: #2166187\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c910f56ec205ee9bd17a14e8787ec42fa1793e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c910f56ec205ee9bd17a14e8787ec42fa1793e4"}]},"branch":"refs/heads/master"},"3c714aa97393a1a18ed80d3af8a8ed1a53e9019f":{"kind":"NO_CHANGE","_number":5,"created":"2026-09-03 11:54:25.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/5"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-09-03 11:51:44.000000000","tz":0},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nCloses-Bug: #2166187\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c714aa97393a1a18ed80d3af8a8ed1a53e9019f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3c714aa97393a1a18ed80d3af8a8ed1a53e9019f"}]},"branch":"refs/heads/master"},"29badd871e305bf5286293fb7f20dd69d6e3f17f":{"kind":"REWORK","_number":6,"created":"2026-09-04 16:29:44.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/6"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-09-04 16:10:59.000000000","tz":0},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved.\n\nCloses-Bug: #2166187\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29badd871e305bf5286293fb7f20dd69d6e3f17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29badd871e305bf5286293fb7f20dd69d6e3f17f"}]},"branch":"refs/heads/master"},"955ed795e51fa6be08f2dd57bac9167bc204b5c2":{"kind":"REWORK","_number":7,"created":"2026-09-16 18:34:56.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/79/993979/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/7"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2026-09-16 18:33:01.000000000","tz":120},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved. The configured value is now validated at setup\ntime against the version-specific vTree snapshot limit (126 for\nPowerFlex 4, 1024 for PowerFlex 5).\n\nCloses-Bug: #2166187\nAssisted-By: Devin (Claude)\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\nSigned-off-by: happystacker \u003cjeanpierre.roquesalane@dell.com\u003e\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/955ed795e51fa6be08f2dd57bac9167bc204b5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/955ed795e51fa6be08f2dd57bac9167bc204b5c2"}]},"branch":"refs/heads/master"},"554d517f40aac45bc149b45511f30d0cee744633":{"kind":"REWORK","_number":8,"created":"2026-09-17 05:42:37.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/79/993979/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/8"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2026-09-17 05:39:03.000000000","tz":120},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved. The configured value is now validated at setup\ntime against the version-specific vTree snapshot limit (126 for\nPowerFlex 4, 1024 for PowerFlex 5).\n\nCloses-Bug: #2166187\nAssisted-By: Devin (Claude)\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\nSigned-off-by: happystacker \u003cjeanpierre.roquesalane@dell.com\u003e\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/554d517f40aac45bc149b45511f30d0cee744633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/554d517f40aac45bc149b45511f30d0cee744633"}]},"branch":"refs/heads/master"},"3edc41259a5cebc6106a2acefdb36b24cbd92e4f":{"kind":"REWORK","_number":9,"created":"2026-09-24 05:41:00.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/9"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-09-24 05:36:04.000000000","tz":0},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved. The configured value is now validated at setup\ntime against the version-specific vTree snapshot limit (126 for\nPowerFlex 4, 1024 for PowerFlex 5).\n\nCloses-Bug: #2166187\nAssisted-By: Devin (Claude)\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\nSigned-off-by: happystacker \u003cjeanpierre.roquesalane@dell.com\u003e\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3edc41259a5cebc6106a2acefdb36b24cbd92e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3edc41259a5cebc6106a2acefdb36b24cbd92e4f"}]},"branch":"refs/heads/master"},"fb5e9608d4a8cd1c2513a4da771e566d9cf17d8b":{"kind":"REWORK","_number":10,"created":"2026-09-25 06:25:12.000000000","uploader":{"_account_id":38286,"name":"Siddharth Kumar","display_name":"Siddharth","email":"siddharth.kumar1@dell.com","username":"siddhvrth"},"ref":"refs/changes/79/993979/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/79/993979/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/79/993979/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/79/993979/10"}}},"commit":{"parents":[{"commit":"a4767351ce74612281497cef7cd7a36f8d784f56","subject":"Merge \"[Hitachi] Update version for Hibiscus release / small doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a4767351ce74612281497cef7cd7a36f8d784f56"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-05-15 05:29:12.000000000","tz":0},"committer":{"name":"Siddharth Kumar","email":"siddharth.kumar1@dell.com","date":"2026-09-25 06:03:45.000000000","tz":0},"subject":"Dell PowerFlex: Fix image cache vTree size counting","message":"Dell PowerFlex: Fix image cache vTree size counting\n\nFix image cache vTree size check to count only direct child volumes\nof the cache volume.\n\nPreviously, the driver used query_vtree_statistics, which counted all\ndescendants in the vTree, including snapshots, grandchildren, and deeper\nnested volumes. This could cause clone creation to be blocked before\nthe cache volume reached the configured direct-child limit.\n\nUse query_vtree_volumes with the ancestorVolumeId filter to count only\nimmediate child volumes of the cache volume.\n\nThe powerflex_max_image_cache_vtree_size configuration option name and\nbehavior are preserved. The configured value is now validated at setup\ntime against the version-specific vTree snapshot limit (126 for\nPowerFlex 4, 1024 for PowerFlex 5).\n\nCloses-Bug: #2166187\nAssisted-By: Devin (Claude)\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\nSigned-off-by: Siddharth Kumar \u003csiddharth.kumar1@dell.com\u003e\nSigned-off-by: happystacker \u003cjeanpierre.roquesalane@dell.com\u003e\nChange-Id: I77291902e477ed14180dbe4692edde6717bbb791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb5e9608d4a8cd1c2513a4da771e566d9cf17d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb5e9608d4a8cd1c2513a4da771e566d9cf17d8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
