)]}'
{"id":"openstack%2Fcinder~995001","triplet_id":"openstack%2Fcinder~master~Ia13b86fb5a16af6bfe1e868f645dbf240ed92832","project":"openstack/cinder","branch":"master","topic":"eventlet-removal","attention_set":{"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2026-09-11 00:21:53.000000000","reason":"Someone else replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-07-24 04:32:21.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-07-10 03:12:53.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2026-08-12 06:01:41.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia13b86fb5a16af6bfe1e868f645dbf240ed92832","subject":"image_utils: use concurrency-aware tpool wrapping","status":"NEW","created":"2026-06-26 05:36:38.000000000","updated":"2026-09-11 01:55:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d4893f05e5602ae5af73778681aa9441718fddef","_number":995001,"virtual_id_number":995001,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 01:45:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 06:07:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-26 06:14:37.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-06-26 06:18:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-26 06:19:23.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-06-26 06:35:36.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-26 07:08:49.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-26 07:21:12.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-06-26 07:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 12:13:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-26 17:08:47.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-03 11:41:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-09 08:25:53.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-07-11 10:31:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-11 21:41:55.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-22 06:42:43.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-07-23 14:53:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-07-24 05:06:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-08-12 06:50:11.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-08-12 06:50:17.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-08-12 08:23:13.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"e44ed2a5d41c18b872e4a307d7b0b15606701586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-06-26 05:36:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b379592e2ca361b6ec9d66fe61aa1ce8360c5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 06:07:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e6f9013a16d4b72bcc4ad2bf08c7fb3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2b9f1f700e2640ef87088142b54a4e40 : FAILURE in 17m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f85fbc4351f44267999b652df2eea64e : FAILURE in 20m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6b3a201fc70941b59a7d6e60e699c3fb : FAILURE in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ceec0ab3e2ac9ab009d7b0c10b5ee221d16385b9","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-06-26 06:14:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/26b11fd217c4400fa12c502cbfd0286b : FAILURE in 37m 28s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"fa91f5d60e01e84fd255c80ad6d811745cd4cc8f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-06-26 06:18:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/a057320392eb47b58623a4c2a52b272f : FAILURE in 41m 35s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/995001\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"5aea3f9187a4eb0e598bc470cb8377a8275cc549","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-26 06:19:23.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/615 : SUCCESS in 0h 42m 25s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"e2115935c3cdc1ed4e96db5ae140e5c5a305e30a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-26 06:20:19.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/342 : SUCCESS in 0h 43m 19s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"a74d3ec15bb394cb5181a80e775bf07273c6768c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-06-26 06:20:37.000000000","message":"Patch Set 1:\n\nHPE AlletraMP Jenkins NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/251 : SUCCESS in 0h 43m 40s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"a2e414ab823839ce9f40d3fb7aa22afcd80fb37d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-26 06:35:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":1},{"id":"11b9f4aaa1dc8078d2544a1b48be5cf4d3274a63","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-26 07:08:49.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/191e80e3dd674b1eba0fa2374eca447d\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1006a485537f4a6d918d4baa1901d0ce : FAILURE in 41m 42s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/522d844a695d4bc88c39956620b4f6ff : FAILURE in 38m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/721d361c229a4ea8b3f819c7b352815a : FAILURE in 41m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d577635de5984c1fa9f0f90c67d1cce6 : FAILURE in 43m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0465a927927a276dca9877ef2d6ebc78917648a6","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-06-26 07:21:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-fc-driver build/4875b43ab6af4633a4eea6479e8c8256 : SUCCESS in 1h 41m 32s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/995001","accounts_in_message":[],"_revision_number":1},{"id":"a3eeb10ec285bbfa1ca7009584cb00e1b63baee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:57:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a68877ed2bdd401aa89f187ee98cf0ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a81f421b5542148407067e89b1ae03 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fa002ea9bb946dc89c58257608911b8 : FAILURE in 6m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42dd3647a9d544a7bc41414da8960781 : FAILURE in 9m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/857d125d46a34729b54d153b96c204e8 : FAILURE in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f020e778e34495ba52a1bc6b6dc82b : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/6df01424f2bd4979aa1774b9e811b39f : SUCCESS in 33m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4e32a8769be437595105e75756a60a1 : SUCCESS in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/508a6d1c71b14a9ca22763cb883976f8 : SUCCESS in 1h 40m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7d384794ba944c69fcf1f04153bae23 : SUCCESS in 51m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec55c3c429594904b66d25aecf6ede87 : SUCCESS in 5m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6727f08cfc94b618197066bc0ee297c : FAILURE in 19m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1ff72adcf2aa49fb84262f7c14b7458e : FAILURE in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0a84c5f256740559700b8f722d9689a : SUCCESS in 3m 42s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/e5859ca6e46d4b02b19cb86495d6ea29 : FAILURE in 11m 24s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2460629203bf40cfa01bf5e7376dc65a : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/350d5ccd68ed4a6f80bb3cd68b26732b : SUCCESS in 6m 20s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/bf92653023ca44fabafbba964e6f8ba0 : SUCCESS in 6m 45s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9c8129d1c463457eaeb391f9ff3d6d89 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47c4405f38e7430ab94fb66839228a04 : SUCCESS in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ca9d123458c440588ee8a5fc7d5dd66 : SUCCESS in 1h 05m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a3e20d5bfff48ec93df05128721e602 : FAILURE in 1h 17m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1bcd94507f444edfb9162a0142ce8ff2 : SUCCESS in 1h 33m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/fb9e7f980e2f4cadb839b1c41ce418c3 : SUCCESS in 2h 15m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8975c7e6658b4adbb95b2e2a69f7880b : SUCCESS in 2h 06m 28s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7b3c23923f0d46fdbb3e362b03b78809 : SUCCESS in 38m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3ded2e4de7c9454e95602cc9b806bafa : SUCCESS in 36m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/51a05452008f44cd8b20a93c9cb4bd22 : SUCCESS in 1h 15m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a74ea3cf41e94861838d578ce0c6da2d : SUCCESS in 25m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b597db6d4c884d1b9c797e8b82862df4 : SUCCESS in 1h 03m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84c5e90da9da4e13ab2f6711efd1c0c6 : SUCCESS in 1h 42m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9aeae5a0bed241608784b35c2d6b0d34 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":1},{"id":"17272f5dc1306149cd6e86d224378817fb4c7c61","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-26 12:13:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/8d9f77e1e91d46f0b61fc6ff5e2ab4d0 : SUCCESS in 51m 37s\n- cinder-storpool-tempest-iscsi-multipath build/4af364b499b844a6869cf4dbfec767f6 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fe89cad08ada69ecb47d95267849ecdcda823319","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-26 17:08:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4543/.","accounts_in_message":[],"_revision_number":1},{"id":"e78d57f5b43c41e354859a040163082636ce63a1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-07-03 11:41:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/995001/1/20260703144058/NEC-ISCSI : SUCCESS in 5h 31m 10s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/01/995001/1/20260703144058/NEC-FC : SUCCESS in 6h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"267a74dbe0d203ef346451581f5521ffde5b8f96","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-09 08:25:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad1704fe82c10eb0c3c3bba9f7eb1240e84fb72a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-10 02:55:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"706cc0a8fbaae6fca4f8d668930524fc8277e9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-10 03:12:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3993e0393496b4b3d886d6059b06099e8b11f6a3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 03:52:41.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8c3d338ea73a4f7282013b2b3d3efba3/","accounts_in_message":[],"_revision_number":2},{"id":"7d74ad45051d4d3b5da3662f9cf4e015bc2121f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 03:58:04.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8c3d338ea73a4f7282013b2b3d3efba3\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7b74baddd3f49aebf1141552314d89f : SUCCESS in 44m 29s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e75126cfc54247aebfd34c908337366a : SUCCESS in 39m 07s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe1d2c10444749b8a3e2c037e73d1e64 : FAILURE in 41m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3245be77946419f9a45f1c981cabea9 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c13ddff16737cdc88273e74228a188e339f0e60c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-10 04:12:09.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":2},{"id":"2f0d5ece76f77ddffcdba4a8ea93f2ae0131b1f5","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 04:16:09.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/567 : FAILURE in 0h 33m 15s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/995001","accounts_in_message":[],"_revision_number":2},{"id":"e2b86cd7e1bdf4053c6fac1918512a324e5dea80","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 04:26:31.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/481 : SUCCESS in 0h 43m 2s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/995001","accounts_in_message":[],"_revision_number":2},{"id":"87b7c950222f9179958fec6a70bb9236de19329d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 04:27:58.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/837 : SUCCESS in 0h 44m 36s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":2},{"id":"d0e0143df2a348b81c6634a870085629ec547364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 04:34:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3889747abc40acb10c5bb84536245d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/025aeb0a89db454bb04ee22748735fdc : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e01574cfd2640ea8d3a3166ad72a299 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e8f0c6a321fc4c628bb5bbad49a8b47e : SUCCESS in 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"896aaf012edf66e88909d0e1d6448c8039010869","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-10 05:00:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/32a199e96a4743ed916097a28c4c04ad : SUCCESS in 56m 48s\n- cinder-storpool-tempest-iscsi-multipath build/90351d1fd2264b42aed864593318b266 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b619c0b9950d43a5cec9c2f581ea3f87a534cfd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 06:20:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/228a254238de4eaf8f9a29b161718bea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70b77e959fc47ea98b21205ed70a399 : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1eb0fe3157634922b423738561e93b54 : SUCCESS in 10m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da9f217f685f403482c2955a879535e3 : SUCCESS in 11m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c3d49d6b44c4eef9d186cbc84220476 : TIMED_OUT in 40m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96bd93d038c643b2a043f49d9f9e7e62 : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/5754f00edb32427590a3df48fdcd6d4e : SUCCESS in 1h 18m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a422bec4581640efb3d5ce0e73f7f0d2 : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5011fcb11739470d9caad4bd468e9677 : SUCCESS in 1h 29m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15551738d7fa4106afe5a6bb22fea9a4 : SUCCESS in 1h 07m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c811ec302164c37858bc98c00c195af : SUCCESS in 8m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/16935a970efa43c1988a9ceb9202e582 : SUCCESS in 18m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/aee1e347e9ec41508e796c23ad2a7241 : FAILURE in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/959d9a906e6547a28178b9d42e3dc34b : SUCCESS in 6m 13s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/fb9acc131ad34d41831a6a6d9f15662f : FAILURE in 7m 39s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2002f1abfb394d6c97713874998c383f : SUCCESS in 8m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d549f9b7c03d4fa2bccb281e142ae5d9 : SUCCESS in 8m 00s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1924f2fa7e9b4137a1ab3fe14c06e8f8 : SUCCESS in 10m 16s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c231f2e53d93485a8b7d0effc120c303 : SUCCESS in 38m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95e081c47c98413e8b1876c4419d64ac : SUCCESS in 3m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8b0723ec9b744cb4ba9090c6e0f5a631 : SUCCESS in 1h 45m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ccd546dc99be4fc39045fd8dda4a11bd : FAILURE in 2h 15m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4039e1780775499c9cd646cf36c14fe5 : SUCCESS in 1h 54m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/28c49588bf1442cea9e2b817a37668e5 : TIMED_OUT in 3h 04m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/2ce526e6db4349888f29728e69f62358 : SUCCESS in 1h 52m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e5f91040589946f5998c716e30fcd066 : SUCCESS in 36m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4f430970a8d242159a7bff5998da1aac : SUCCESS in 40m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bbe141c4554b4de0a883f8599dffdeff : SUCCESS in 1h 36m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38233b4f67f4468bb7cf921fa3c67bdc : SUCCESS in 43m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d568bf21d6ca4db7854b582d3fd6ad1e : FAILURE in 1h 22m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f50275f79d7403f89686be5f6765082 : SUCCESS in 59m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0491bb91614941979fdf3b5d7077920e : SUCCESS in 31m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"da301850ada68b9381ba7d0fcea5575f5889f7b0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 11:21:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4759/.","accounts_in_message":[],"_revision_number":2},{"id":"4ea9e84f69ae62810995a7f9e4403317952dcd5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 01:19:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a6fcb76a58f68f95d4b5a72629ebadfffa3c0f8b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 01:22:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd2419dcd29f41437527fbc9f16d89f24898accf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 01:37:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540455fffd294c97b729f4db176ba456\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/55f270b9ef1a416394305c7721c70b33 : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f4f4acfab544be8a2d2ef8e981a7f16 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c2d6c162e944310a5c554a53e5498ee : SUCCESS in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7da83af0400534a7e55860625a40bb2cc309dd27","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-11 02:02:29.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b94dca022fc64d69b949f67d1d28cdca/","accounts_in_message":[],"_revision_number":3},{"id":"72e8535f722142f2804d1320fe1fe24a11c1983d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 02:05:01.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/495 : SUCCESS in 0h 42m 39s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/995001","accounts_in_message":[],"_revision_number":3},{"id":"f3f0b4c259b5ef1b6c167346b9c2f524998db1ed","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 02:05:10.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/851 : SUCCESS in 0h 42m 54s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":3},{"id":"6784dc997a8b09cf69fc2f66ae7aad9b538c0352","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 02:05:44.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/581 : SUCCESS in 0h 43m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/995001","accounts_in_message":[],"_revision_number":3},{"id":"3c71f894a508b98c9285005460bb9f9a580ead89","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-11 02:20:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/42236a91d662457883a90934739b3734 : SUCCESS in 52m 31s\n- cinder-storpool-tempest-iscsi-multipath build/db1eeb1354a741e98a5db41829275e61 : FAILURE in 1h 00m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f9c05828a6f8815c9076ae0e5f2649c8e073f1ce","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-11 02:22:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":3},{"id":"7c3ab5bd9a658af9f651aceee10be37192cf4892","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-11 02:35:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b94dca022fc64d69b949f67d1d28cdca\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2af30f6d5d52433789324afc1084320f : SUCCESS in 54m 39s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de73c2fd9abf4194884b5b5f65b71009 : SUCCESS in 39m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2440106af84489f950d8e9b3a6b69f6 : SUCCESS in 41m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d7317e2a83f441eb570ba9fab79f275 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9991d567010a36afdfe8db19e27e5240cbd1f306","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-11 03:43:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4773/.","accounts_in_message":[],"_revision_number":3},{"id":"ac479dcf4155a331dbad84595734fe9c6dae665e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 04:31:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d45b75f80c484b12bb33f731f88176d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1156426a4fa64400b5b6b51ec2c0d2e3 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33fe9d9b8061439eb9c2b15bb800ab50 : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfe56f395a5a486eb7af58c1073fd817 : SUCCESS in 12m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee158af7144c4b478ca3f7ff989950b7 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37820582c781461aa2ede3ea9db2323e : SUCCESS in 14m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/dc588da2f85048e881c0fbc770d452dc : SUCCESS in 49m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc6210876e3042de9c555c665fb1223c : SUCCESS in 1h 09m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b2449ccbac64ec8bc1ffc52c54aa606 : SUCCESS in 1h 39m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0d121f8f99341ab8c25ead1d939020e : SUCCESS in 49m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52e9a5ace5f6402eb2fa57cae440b440 : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c0e6cd178bc4f159a10f03994b395f9 : TIMED_OUT in 41m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0aba12ba53a4da5926958244355f6fb : FAILURE in 5m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d49ca3aac884f5089547a56819a8417 : SUCCESS in 6m 09s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/424bc7bc30474a97835bc7454b9faae9 : FAILURE in 10m 16s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fbfee59ce5df40f28d1b345b7a9578c1 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dff2617007004066a143e9d014cdc6ea : SUCCESS in 5m 39s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1681cb83b3294e0eb43781c5bbedd3a6 : SUCCESS in 9m 01s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ea3046021734e9cbaf0c66cb35e5a7e : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae8d11e72e4d4a098fd4df2f5c96572c : SUCCESS in 7m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2daf196058b4f8fa1ab8099a5598298 : SUCCESS in 1h 52m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2186eec7138148098780f73ca651d38f : SUCCESS in 2h 11m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bfa21eeb8cd84127af2d796dae65fc6e : SUCCESS in 2h 00m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e9b8c288bbc64962be08cd3edbbff474 : TIMED_OUT in 3h 06m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e0c2b1736a2b4eccb16d23deb0b8d67a : SUCCESS in 2h 10m 58s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ba57f41259aa4af9bde0ae59daf99708 : SUCCESS in 38m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5498934d62284e2396ac5aebae0d9c90 : SUCCESS in 1h 15m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f751e06ab41429298e0acc13e7b20c0 : SUCCESS in 1h 49m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2bd20d88c7342fa82cbbba61ab10c90 : SUCCESS in 29m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0c0d600bfa7149eea16fe312e280202c : FAILURE in 1h 48m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c78c421dd0e84e99b0f4aff2775fe1e4 : SUCCESS in 52m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d7a2f95e9cf44fe8aeed102a56a2052 : SUCCESS in 1h 03m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"eaa163085c529c8056f8baef44529b9b49819dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 08:09:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5845197a96d02529bdf64527ad30c4b0f1df8b1c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-11 08:50:19.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0328245f2d9b48d89270e42e7667eacb/","accounts_in_message":[],"_revision_number":4},{"id":"b318afea8182a07af820a1bc25f7e369ae9f5fd7","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 08:52:39.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/852 : SUCCESS in 0h 42m 55s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":4},{"id":"521af88e1ea309bbed6129884bf66f9d4edbb4be","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 08:52:58.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/496 : SUCCESS in 0h 43m 16s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/995001","accounts_in_message":[],"_revision_number":4},{"id":"66167cf75b4047ca43b9f0abc2dd88a259dda8e4","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-11 08:53:36.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/582 : SUCCESS in 0h 43m 51s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/995001","accounts_in_message":[],"_revision_number":4},{"id":"355fee129cf5b290b87d68faf0d66de46ab44d98","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-11 09:06:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4db06a44b0bc4c8ca5e3e039bed04ae4 : SUCCESS in 51m 34s\n- cinder-storpool-tempest-iscsi-multipath build/5720d23d75df4efda67f2d6dfc6696e8 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":4},{"id":"4e84b5f93fbb7059273444d523a7c0bf1c757608","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-11 09:08:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":4},{"id":"be5332aa827be7a5da5ea5f688dd7eefff3945d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-11 09:21:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4774/.","accounts_in_message":[],"_revision_number":4},{"id":"84e6fb58ccbc9166c978198c803ea2367e3fd537","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-11 09:49:16.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0328245f2d9b48d89270e42e7667eacb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f8d3b6011ab94738a367c0615982ff2d : SUCCESS in 44m 48s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5cadf105d6f24c749feb092668ac5286 : FAILURE in 1h 39m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/72892da744c944e8b45d076815645bc6 : SUCCESS in 40m 18s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/412fe732d3f64f789d0eb1eb8153ebb7 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":4},{"id":"21bd57b16ec065a8672444bb3c3d222b99e39386","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-11 10:31:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-995001-4/FC/ : SUCCESS in 2h 21m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-01-995001-4/iSCSI/ : SUCCESS in 1h 28m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"555ab7258013113bd71f0b46d51d10b8dc856d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 10:57:23.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dda6553aeea4a3b9c2656a37b6f3841\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade5c33fb3e8408fbb5c4bd0203f006f : FAILURE in 8m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93b373d616b04f629267f7bb92cb776a : SUCCESS in 12m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba6475cfa3fa43729b51cadd5d93bd3c : SUCCESS in 11m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/764ef7bd32694792a306e82100f1edf5 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339f5f4713a843298b512515c277a185 : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a4ff5676aaef45dfa77f49109de9404f : SUCCESS in 49m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/215d19d0c2084e3e938d99b26533acea : SUCCESS in 34m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ab47d02fd34c008ff2d55827dd6cad : SUCCESS in 1h 42m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dedd249b2d4d47cea3dad6f95dc80fd7 : SUCCESS in 38m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c386a21526c437eaeee760b2d8ab910 : SUCCESS in 5m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3736aede4d6b45269e6a75d7e7c56207 : TIMED_OUT in 41m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4fc183b9008a4a7299e7ef2251e30d20 : FAILURE in 6m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6a3e6bf88b247c9a8916b37071c82c7 : SUCCESS in 5m 59s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/6cc31d59b7774f5fa1424a2ecca5278f : FAILURE in 8m 43s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/24fc68fa71ec4dd1ab1658ebc64f1e85 : SUCCESS in 7m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/064ef266370b42ac884c1a0e4481b08c : SUCCESS in 6m 56s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/e4c3192b1189434f8cf1601288c1dadb : SUCCESS in 8m 29s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca5f53c138484db09f2b0f3db0650761 : SUCCESS in 41m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5acf2e7e774543b7a45fe988e3cd7194 : SUCCESS in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/359f6079fe4f424982748e4dff0072c4 : SUCCESS in 1h 44m 49s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31c62a71fdb74b33b239b39f92c07e88 : FAILURE in 2h 29m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c2561311e94f403f8b48225090327792 : SUCCESS in 2h 06m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/89bde415b104431ea597d209f163eea3 : SUCCESS in 1h 44m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/13216be1c8c74eef8a5724657de0c91c : SUCCESS in 2h 38m 19s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f2e3be6109da472899546e2316b4f1d5 : SUCCESS in 35m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fba8e378489a452d97defa12429d17b8 : SUCCESS in 1h 13m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/70b0a080ec3447e78e8498ed3a754af6 : SUCCESS in 1h 46m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6c73121c0a204c0c9d4109cf911fd872 : SUCCESS in 28m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/31d088e31b9d4f32a840af4b622aee97 : SUCCESS in 1h 15m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/260aa88c3f1d4f88a3ebe21b36c5eed5 : SUCCESS in 1h 33m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56740f0504234d78be75d7fbad2ae372 : SUCCESS in 1h 01m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"ac8e8fc6c6a052cef4f9ccdd363ad4da06b6af1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 11:06:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74cc1cf18d3f4f4d84cad58b23d19666\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5719e3f3b26c4fe6a7f687c9e23d69d7 : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/965b1e20aeba4ae8874f365a5d0f5112 : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/afedfd06fda5479d99d088cea6557857 : TIMED_OUT in 1h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"28a4125cd30125558b444713a2d3b57e46511cf5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-11 21:41:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-4-fc : SUCCESS in 1h 13m 2s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-4-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"29620d43347c92dccaf5f5cb0d1b0f6a0adc1604","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-15 18:47:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4367/.","accounts_in_message":[],"_revision_number":1},{"id":"ed0c42146ec8e6419c3e6ec10fe251925b8e5323","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 22:21:33.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4583/.","accounts_in_message":[],"_revision_number":2},{"id":"8fefa101d89add06fb8a80046dab87ddbbcc5258","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 01:15:25.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4597/.","accounts_in_message":[],"_revision_number":3},{"id":"dc87afb808da7d9fe1a9771f79a48acdbd630ebe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 01:23:44.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4598/.","accounts_in_message":[],"_revision_number":4},{"id":"261fc3094505aeed6e92a0af5b3800a0d6e3b111","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-07-22 06:42:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d8306d672ab014d83026e9a70d5470b5ef351280","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-07-23 14:53:47.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd9a9c090c6051103fd0ee289113799dd762ee70","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-24 04:24:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"efaa90c94b8777607ceb1008df8ddfa4501304f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-24 04:32:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"22f65a21d6a9347e691b53a639b5ff2343d2cbef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 04:41:06.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4910/.","accounts_in_message":[],"_revision_number":5},{"id":"5409d13fa4eed43526473387375c6a97e93f3905","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 04:46:08.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4734/.","accounts_in_message":[],"_revision_number":5},{"id":"697123da4a1c28cfaacedf8c7001da97d8072312","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 04:51:09.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/728fa13f240549e29b12829869062d96/","accounts_in_message":[],"_revision_number":5},{"id":"f734e6709d1cb9c1da41449bf8fbc061150ee31e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 04:55:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500965bd1cfe4d8eabe277231416a81c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/271c14ef40f9464a9f86ae3a391a6ba2 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3db7d01c1ee348b5a6bdb0621d18729b : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1d8f6a8029a74379a7a8533becd5830a : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f82413a47a10e6dcd9f2b4a6b1609d7ed6005981","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-24 05:06:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/995001 : FAILURE in 30m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2200bcbbf638bc05e81938df05f335f746cf0244","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 05:18:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/728fa13f240549e29b12829869062d96\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a1e6806a173e4831a79040d698d6f26b : SUCCESS in 42m 04s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/24f62d3408cf4061b64929531097d975 : SUCCESS in 35m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/573c0778fb4e4a48bdcc397587fbbdae : SUCCESS in 45m 52s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8ab9917d2db4c58ba30a9f4ca8a0435 : SUCCESS in 34m 45s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e324318441244b0e80d5e10fccd48955 : FAILURE in 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0760c5fd73fd7f3c127c5c894d323b970d2a3525","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 05:26:09.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":5},{"id":"e1536c7e319c7f4989a6aef3ed8fa6e4935ced65","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-24 05:29:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0f513986f65f4c749dca89be7eb73ea5 : SUCCESS in 49m 47s\n- cinder-storpool-tempest-iscsi-multipath build/76abf41dc6e643409bd4e0b1a9ee5061 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":5},{"id":"36eda89b204cf0749985765563205c196cb3431a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 06:38:33.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a23f01d15424b038bdb7fb8859a7a7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e343f3c67a544eb2886881c69da0f0b0 : FAILURE in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ac82192e64043a88656a3bb34c4b261 : SUCCESS in 13m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ac8d379a5624b5a811f444f86d8c097 : SUCCESS in 9m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f385f172dba24a60b4dc07484393d2f9 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20409965bfe4656806cfb4e4ad40a64 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/5fec41e71fe84b8a9f1991804fef733b : SUCCESS in 59m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d313331472c64a5c9a3351807e2f6db9 : SUCCESS in 43m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98465ef715f84da686cdb9bcfcc89f0d : SUCCESS in 1h 59m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe97e47d04c14b20be1ee68c9b258eb7 : SUCCESS in 1h 03m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b60e0c5fa8734827ad0f419334f10220 : SUCCESS in 9m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c6f9216d46d474cbf4940e43aabe86b : TIMED_OUT in 41m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be89201e79574fa6af90dc909b33ed8d : FAILURE in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b349d929dfac4c0f95a74ce0953c4e73 : SUCCESS in 7m 14s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/d6688bb0b9b44696a3aa95c3587f3025 : FAILURE in 10m 19s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/492076850d98480485da9d677eb4326b : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f91932235850482ab6ef4f2fb73c0a21 : SUCCESS in 5m 50s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/02bb1b20982c49818b05608c4609121e : SUCCESS in 9m 11s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12be32ff6ba242d79f3241b6d3892804 : FAILURE in 28m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/10160285b06a42bd99e731098bd2c5f5 : SUCCESS in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57f9b21470164f77a30eafad0d28fac9 : SUCCESS in 1h 20m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9ff46c483db5498eb0f6c77b5211d252 : SUCCESS in 1h 44m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ff44e43d67049598b6f4e99df0075c2 : SUCCESS in 1h 48m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ecbdc3018cfa452d9dc291194567713f : SUCCESS in 1h 39m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/f79a0e409ed24de4856376369e3e5ea2 : FAILURE in 1h 25m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/4a9baf57b505408f904e574002d814f3 : SUCCESS in 22m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cc95740604ac4b9c89ba78054a8bc919 : SUCCESS in 46m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e9d6b17f358d4e5198a148399840637d : SUCCESS in 1h 20m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f3084a1c96d54783bedec4ef8e285bd8 : SUCCESS in 32m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/14fd0ac7f7e045448cf4ad6e83784cbd : SUCCESS in 1h 06m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/783153fa45bb47ea98e7e18b20cd2ba4 : SUCCESS in 54m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/230c5a1594c5428e8b9de230a9cf3d02 : SUCCESS in 45m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"a31268ccb867b6c91071d6b8ecf75a6953a73695","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-07-29 13:26:58.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f0f8fc5d790ef563af619f7ac31355eb204d7e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-08-12 06:01:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"23adb5882b2362d85bb368cd4cf0ac6b573a5d7c","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-12 06:11:06.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1151 : FAILURE in 0h 9m 4s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/995001","accounts_in_message":[],"_revision_number":6},{"id":"9933ea289a9e062734b186c611ec5762b9714359","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-08-12 06:19:21.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ecdb9c5c5da6ca1af73708b54daa573f36df6271","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-12 06:27:56.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/67675fa1009c427b98a9e8aa955f9ab9 : RETRY_LIMIT in 24m 40s\n- cinder-storpool-tempest-iscsi-multipath build/82d9053bcdf04f41b4ca3b6a76727909 : RETRY_LIMIT in 24m 41s","accounts_in_message":[],"_revision_number":6},{"id":"f8096d5aa2b0902d2ef7bfc0e52bcbb1cd67fc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 06:28:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc7d15e25a442faafbbcbcc28ff5d3c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18ebc5486cc84f87b46c722778b78984 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50e095899c8e40d6bb7fe8ddcd57f491 : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f8cf994295f44b7ba19e35b069a81925 : SUCCESS in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"58d5a3eab20f6f9cea0e3e744b4850b4df30baf7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 06:42:34.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4386d7b48a0d4d2b984dd78f4994857b/","accounts_in_message":[],"_revision_number":6},{"id":"3714c37f884da80b0c20070e978876e3ef53a62b","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-12 06:50:11.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/790 : FAILURE in 0h 48m 5s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/995001","accounts_in_message":[],"_revision_number":6},{"id":"7bb5094317c76b75da15d1b963a66c300e3956dc","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-12 06:50:17.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/884 : FAILURE in 0h 48m 14s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/995001","accounts_in_message":[],"_revision_number":6},{"id":"172880905f0940f9be468c85a4ccd6e8b4ea2923","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 06:58:50.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":6},{"id":"d29ca9c5b87178b178ec44b1f462a163237c3dbb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-12 07:27:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4386d7b48a0d4d2b984dd78f4994857b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3c112c78130c414191800e75963b0d68 : SUCCESS in 44m 22s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/604374693822453bb81d16c309cf0a2f : SUCCESS in 40m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/597d616ea8974c248a92c43c58a6f13c : SUCCESS in 47m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6b348210eda4d01bf0f0bb3e112cba4 : SUCCESS in 40m 13s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/00b2011d0e5c47c8a8bae48f0462f1f0 : FAILURE in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3a55bf6f14a2338755a7f45f0bef412dd90477c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 07:59:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4920/.","accounts_in_message":[],"_revision_number":6},{"id":"5a5852538d8f9451a302ce90942128387e936a64","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-12 08:00:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5096/.","accounts_in_message":[],"_revision_number":6},{"id":"3f28da6e360c890a36c4ac424086c34d52d002c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 08:23:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e22d2f63af54eab939e01dd3e276365\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f1bb470d6e44ecb7172e622fcf093f : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d99dacd9645e405d910b936a2f5f86de : SUCCESS in 10m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/088d51cf3b134fa4a624f5f6cb8fa2c6 : SUCCESS in 7m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/806913d42cb14a2ca139372ef7ee2283 : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2187f49b8ae2446e95acdf1f0ddd0fca : SUCCESS in 10m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1c5a2691f97b44bfbc5d195061139a42 : SUCCESS in 1h 02m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e158e6eb47c84e04805ea57d4b66fd50 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/636c69c8cf9343c79568a83fd29ffcd4 : SUCCESS in 1h 30m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51684633a381405bb782a74fce764ca3 : SUCCESS in 58m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/846f21375245434cbf1853fd6ea6b3d1 : SUCCESS in 6m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d15c3532a3ce433d94f638eb1931629d : SUCCESS in 16m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/491f79fe7b44482a96e6f349a9acf29c : SUCCESS in 4m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cd7dfb09ef6498cb1b29440f121042e : SUCCESS in 5m 21s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/51a47c93887c40758f40234fba3ae0d3 : FAILURE in 11m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6f42c6c019b48119c77e117b7352c2e : SUCCESS in 9m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/829a3e34e5eb429a8b2d13771a3509e7 : SUCCESS in 5m 46s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/99130eb0f7da40bf95743eed329f4281 : SUCCESS in 6m 24s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d4d5c6a25ca46c4b1a5a7981533f369 : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a33bfd898804c6b8e694e601f380377 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/91e22e0676df4bf39c2d2585f903f2b7 : SUCCESS in 1h 29m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ed160fe17791444a8c92e558d95d1238 : SUCCESS in 2h 00m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1822d402c3bc4c70b2a1fafa04420eeb : SUCCESS in 1h 34m 13s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/5e1387d2783d473a9069a4a4588228e2 : SUCCESS in 1h 24m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e65c90e03b504924bab328c73f6240fa : SUCCESS in 2h 15m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/38fa31beb54e4b13a237f1d5de6df308 : SUCCESS in 18m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58d9390bd54e4855a79ae4309576d610 : SUCCESS in 1h 01m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5ef0e2dfc1ac449e85e4d255d4c94eec : SUCCESS in 1h 27m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a6a29e4546b4fb1a6f4d5e640d1279f : SUCCESS in 27m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/457c113dce6d4cbd9a6c8adc98fe4879 : SUCCESS in 1h 48m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1568cfbdabee4cc7bc17da75a87041a4 : SUCCESS in 1h 01m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97b4f8328fb84dbda67b34a7ff6309ec : SUCCESS in 26m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"8e59d8ef06c6d40ff3bdc26bcf9239ac66532f8b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-12 08:33:50.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_08_2026_08_26_49\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"a3252c646320e7f046ff3e1e14f7dc2d50b84694","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-12 14:35:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-6-fc : FAILURE in 18m 42s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-6-iscsi : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":6},{"id":"a5a6214dcb889f7449e9fdeaeef82de21ae1359c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-09-03 20:21:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"8fc6f1c5209b73f78b865b83e83cdc5ccebfb73f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 20:39:53.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8b1ced2be753417fa15082901a9960bd/","accounts_in_message":[],"_revision_number":7},{"id":"df77ec8888b82407b73411cbfa17dd3dc20ec0db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-03 20:41:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest build/0a2cd3c582134698bfdd1e7184db82de : RETRY_LIMIT in 17m 48s\n- cinder-storpool-tempest-iscsi-multipath build/eeb6fc0d854749faa034d7d007589233 : RETRY_LIMIT in 18m 29s","accounts_in_message":[],"_revision_number":7},{"id":"5b43410620ea0a4af600d82f4d0ae35dfba3b8f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 20:45:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e532f2fc5f684705a6eea37de8581811\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6fff072132b34e6289858abb86ecaef9 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32401f24964640ecb8bc46ba4435d707 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8624df43a52e4b1dbad535711feb08a2 : SUCCESS in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e74ec72b271ed791fb64de014908834ab20a1928","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 21:08:05.000000000","message":"Patch Set 7:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1076 : SUCCESS in 0h 46m 36s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/995001","accounts_in_message":[],"_revision_number":7},{"id":"29c5313b7363a6e42657a6d9e0170d3bb5028920","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-03 21:08:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8b1ced2be753417fa15082901a9960bd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5cac437f9a4347089df7d1c2ea456aac : SUCCESS in 47m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/966cc3476d9b4e04b719cdb0988da3cc : SUCCESS in 36m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e1e7406e50048c697267ace94dd7ea7 : SUCCESS in 46m 37s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2fd8f3d782f74b55b8627b5ae9c18f7c : SUCCESS in 37m 20s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43878d74a5db48a39f8a41d692837780 : FAILURE in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0bf09777533767ecf1a9e91f0c93bd417fe55d1c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 21:08:49.000000000","message":"Patch Set 7:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1437 : SUCCESS in 0h 47m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/995001","accounts_in_message":[],"_revision_number":7},{"id":"92c86c061e7bd924bb05ef781f87efc2bc9e6177","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-03 21:10:33.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1169 : SUCCESS in 0h 49m 3s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/995001","accounts_in_message":[],"_revision_number":7},{"id":"538e064956a340265c3444e134488daaee3e75fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 21:16:21.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":7},{"id":"db90aec8329c13a5d24d2b278ada83ab969a0464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:28:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1442ec5e634aeeba33a96148a445d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c066841e6c474f83f5d605b399276e : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/836e2eb4592f4644b09e94d5f25f5014 : SUCCESS in 11m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32435044bb994839952acfce395d93ab : SUCCESS in 8m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a13640ea2e254411ac3d4eaff03d007d : TIMED_OUT in 40m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224d425c2f084da59cb844e991bbc265 : SUCCESS in 13m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/cc20774df8664843b2a0208257f2f97f : SUCCESS in 1h 01m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1df3ac0862014eb7aeacc7c724aeda2a : SUCCESS in 1h 00m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c001d5063f1407093f28059e95619f6 : SUCCESS in 1h 03m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc4a5bba94114c05a6440e76b2a03138 : SUCCESS in 56m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83b40b0c0fea429d926e8ec8f2d6e1e3 : SUCCESS in 9m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/83e2a96ee3eb4b539c3d91af73eb20bc : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b0df0751657e44d2a74b9d7b9cb21f1a : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3744c20b8d8431c9b50942e4d20e9dc : SUCCESS in 3m 48s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/820571980850423cbe1d195c2e4d5f84 : SUCCESS in 10m 12s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4d1c47f16391438d910c7261d3d77a1b : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68798e4bcc2f4e908b5604c782353935 : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6a82b08f08294d94894be543abebd492 : SUCCESS in 5m 43s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e7615afb5b6640e7a3cb9a34f6adb238 : FAILURE in 4m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a7e4d3da3bff4efdbc825f266f2261ba : SUCCESS in 8m 59s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2d6fafc3c0bf490c9103db04b783adaf : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ff269c4f28d45169a62202e4e8d56ed : FAILURE in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b5b5fd220f4437599a671e1a6d12131 : SUCCESS in 49m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be2f5e0416ad4c5489bdd4e9de574982 : FAILURE in 2h 01m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6d806f685f1342f49b71d294f788d02e : FAILURE in 2h 01m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/83f222388e644b9fa230be0249f15cc0 : SUCCESS in 1h 23m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3de8254234724c6fa205a4dc07da5e46 : SUCCESS in 1h 52m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d649faa38d5248e999742f77dbacf284 : SUCCESS in 40m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1b3f0c3acb548e092829c83ce781e79 : SUCCESS in 37m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b42b76af90b04f6f9685a1b94c90d513 : SUCCESS in 1h 12m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a82bf84e30b7452a9cc223e77d27aaa0 : SUCCESS in 34m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3d0f7672aba473895bddf31f1461b76 : SUCCESS in 58m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e97cdafd78b341178b0d8a20ff5a244e : SUCCESS in 51m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2fc1d1e55bb439ab3479bdb95ac2fdd : SUCCESS in 55m 27s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"c4ef7dcf2339e54ca7e584d8a7d1021f794ee63d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-04 00:07:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5404/.","accounts_in_message":[],"_revision_number":7},{"id":"79c3d5475123bfcfc5db4d005469e34693c01db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"real_author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-09-09 19:08:00.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_10273\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"_revision_number":8},{"id":"a093cd09db2b47d89c51eed3d49e92d1dc3bd84a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 19:25:11.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5495/.","accounts_in_message":[],"_revision_number":8},{"id":"a8012fb356c9458e18418bdc12f301c5e740484e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-09 19:25:31.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5318/.","accounts_in_message":[],"_revision_number":8},{"id":"1513ea7c30472a2a5007699d93ef588628a17c04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 19:34:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bec82e6fc1640678a46c091ee6a9dfe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/938c4999fb204e5eb73b1fdd8d6dd8cf : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f82134d360db42f4b2a226b628d5df3b : SUCCESS in 18m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/448600d0836a4ef0ab2650308e9379f0 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"54b5ef27fbf49704f2f7e2a3d1b588bb9eb7744f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 19:40:18.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/eb14e82b067b449989630dd612eb33b9/","accounts_in_message":[],"_revision_number":8},{"id":"d83b519e2b00c76ac83a08e907490ac89517f40f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-09 19:41:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/6d08bfdc2af54e9babb06e785b655fff : RETRY_LIMIT in 16m 47s\n- cinder-storpool-tempest-iscsi-multipath build/23645887dac84361b48cdaa8ebd61883 : RETRY_LIMIT in 17m 30s","accounts_in_message":[],"_revision_number":8},{"id":"4e8c7cdb644f1b09402541818b0023849735b21a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 20:02:02.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia13b86fb5a16af6bfe1e868f645dbf240ed92832 .","accounts_in_message":[],"_revision_number":8},{"id":"ab0b27dfb30cd53789bb66fe16faba5e07eedcb0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-09 20:50:36.000000000","message":"Patch Set 8:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 51s\n- FC       : SUCCESS in 0h 47m 34s\n- NVMe/TCP : SUCCESS in 0h 48m 40s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/995001","accounts_in_message":[],"_revision_number":8},{"id":"798457e6e3cc0a2bf3fb21eeac22bd4480e00714","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-09 20:58:21.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/eb14e82b067b449989630dd612eb33b9\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0dc22101d15d4562842d3c3ea6931bcd : SUCCESS in 44m 11s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d400cd5f2d54dd58897051fe38fa49c : SUCCESS in 41m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7b1fca8efd14417382e472b1b1dd0870 : SUCCESS in 49m 51s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/52e9a50c0ebd48cba6c5fd5393342d90 : SUCCESS in 31m 48s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/89e002ab46e34573974381ddceac4352 : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"12f1a33be668f9afdec54593ea19ed529b541f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 21:23:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dea0bbf648c4397be166351545323f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc33cccdd3fb4572bae941df86047d49 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d6f1f128413495e96c511b7bb9d867f : SUCCESS in 10m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3877fde7dbb74f97adee87070ecca495 : SUCCESS in 8m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98095e4be2144f2cb8c9981f6838d8e6 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f434d79b3db4d0ba9974a8f3197f60d : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b9c099a3b2b94dacbe46364fde3e0d79 : SUCCESS in 59m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/47d881832efe4bb8b1f81532ba05fc09 : SUCCESS in 32m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b38b3710cefd4710a97500854b30c8fc : SUCCESS in 1h 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6fd97186e6a42dc9fba58b09ed7e625 : SUCCESS in 46m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c67fe2145e14849bcdee295f82d96d7 : SUCCESS in 9m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d18236ff04d54b78af00bbbf0f1f6ad9 : SUCCESS in 16m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac7fbf7371094213931bd4978b22c59a : SUCCESS in 3m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/66fcc40e5b0f4ed5a1c4776949f5d660 : SUCCESS in 6m 31s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2338c9cfdd4f43f78457b22deb4dcdaa : SUCCESS in 8m 55s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/17ae9dde419d4b06a3f6ceaa70231d96 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/945acac106254686ac8c24cf905ee0e9 : SUCCESS in 8m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b0456b43e2ef47f0bab13292bb3a88b6 : SUCCESS in 5m 42s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/de14e62ce4f64921808b78bab41476f3 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4d1c49a0db0c4744bcc227171a9c3679 : SUCCESS in 11m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/979bc57b88ea4f59adb876a07bccf0a2 : SUCCESS in 36m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2975f653cbdf4db98a0af8017a3134eb : FAILURE in 3m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/77e725a543204fe4987572eedc7872bb : SUCCESS in 1h 37m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f916a1b63aeb4bc788b2842ddcf4a5c8 : FAILURE in 2h 07m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc26a57e401b4a69ad38c966a3dd3b38 : SUCCESS in 1h 47m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ffa1f0fcf1964eea94a9302518c2cbbe : SUCCESS in 1h 28m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/cc1378dca8104b5eaf4083171c3725c2 : SUCCESS in 1h 34m 12s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/49f6b5111ad2493694b05e8789fc8c8d : SUCCESS in 37m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4c51d2e77d88413c9849548f9a41815c : SUCCESS in 1h 20m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1007ffae34724bfb8df82cbc3de79117 : SUCCESS in 1h 45m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/db5f9c33fec246109ba307c4f708c6cf : SUCCESS in 26m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/515629a5a6c14140b16944274ba0076f : SUCCESS in 1h 00m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4cde08af0b54982864828673a69fd73 : FAILURE in 1h 30m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60d799a5e89f46e2b343e62f7a8d8803 : SUCCESS in 43m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"ee5e4f5e1af48c5b5d916d490e8cfb52ba03335c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-09-10 03:52:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"02cdef35d8b1be1be4b6da3df7def16cb7329df7","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 04:03:19.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-8-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-8-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"d8df019fd17fec5c5ff1794da87aa3cc36593dc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:20:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18871c86aa06454c89625519a604417b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f94a2e4ffaae4878bab7f99da5e1ac9d : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ced16b74ca804a3d838ca499d70cbc6b : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/01b332fed7af4e729720935e365849d3 : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e4988f6c8ed4f2fbc3a478cbe66709c3f7a3197a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 04:28:39.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4e1a549ce74e4b768136d86cd44c495f/","accounts_in_message":[],"_revision_number":8},{"id":"58dea263d96cf4fe2d64b5f1584d6e95b3582af2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-10 04:42:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4e1a549ce74e4b768136d86cd44c495f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2c3f0577c4a6473db2a4337c1158e595 : SUCCESS in 40m 05s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff6fb99747254cdb8a31e69990613231 : SUCCESS in 37m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9c9582265bdb464298fcca56d6c6dccd : SUCCESS in 46m 27s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e9870bb8d5849b0878576f05a51511b : SUCCESS in 35m 42s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/86233c85a42140f188373f3aa143242e : FAILURE in 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"083b9b6f68641ed06d222b19f9e5af6f4465f49d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-10 04:44:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/287ee76dee664770a497590ba714dc44 : SUCCESS in 45m 32s\n- cinder-storpool-tempest-iscsi-multipath build/25046d6a2d6d4bd9a9aee8b1de0d76ab : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":8},{"id":"c670f243d548577152ea285d02f1f203090254a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 05:54:10.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/077f0c9b09b3496498b257d4aad2bfe0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/257d10a4917749cab46e1084423a9dd6 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc9b7a26292a43b8b66c3437e9d40ef6 : TIMED_OUT in 41m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/473572b8897b4d7c81f746836d839535 : SUCCESS in 8m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e6180952a10443c980a29028a908bb9 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207429143bd94bd19e87bd586354263e : SUCCESS in 13m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/16e50f1305a747c888d76a4db8834658 : SUCCESS in 1h 04m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3846254e2a5b4871998ab887326504d1 : SUCCESS in 37m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/400522a3fc744196993a83f29c5f4994 : SUCCESS in 1h 26m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d0c9c1f9cca40efb64d5aae97af15f4 : SUCCESS in 40m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62b3dd46dfc14a06a8074948f51a4b6b : SUCCESS in 10m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d323de07b85445f8ba1461885fba275 : FAILURE in 23m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9915892176be4533bb4d28138f675057 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c3aa63a151d486a87f321b3a3d255b5 : SUCCESS in 3m 57s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f0f908f68b2140af856e3f47f91950d3 : SUCCESS in 8m 22s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2a745d53c558430d986fc8b2f6f3dd19 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/af369fe3c8af499b865e1ff534a884a1 : SUCCESS in 5m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/18f7c906ce274b0c898257d2ed6629b3 : SUCCESS in 5m 59s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d0528ff53ee5441fa242cd163be86b96 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9dffe69023e54e8d8e544233fc483bc5 : SUCCESS in 7m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d6a0d11c3bd49dc9653f67264fd3a9f : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bad707b805a34978af981d9e47092e8f : FAILURE in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/34b260e2e66844fd955ebd585eee4e7e : SUCCESS in 45m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ec607270e2fa4faf91cc55c0296e072c : FAILURE in 1h 56m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47643d1aa914431bb81d9b599af4b46d : SUCCESS in 1h 47m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ea0d7915bd7c4e09aa9139e73e57aec3 : SUCCESS in 1h 36m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/15442c6ff9be4e0a813ab70d7b9c776f : SUCCESS in 1h 26m 02s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/9b1a01d6c49e4d6a9dd8e89713c410ea : SUCCESS in 21m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b6a86ba6620643f0bdc44187ea5d0509 : SUCCESS in 44m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d535c7cb6ed44aff8a1b33e6a57ed9af : SUCCESS in 1h 29m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d411d15c9914d2cab962ade71f0b6ab : SUCCESS in 41m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f5f554758ace47758abfbfdeca76a5bd : SUCCESS in 1h 41m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/982d6c67b6ab41a8a13f89c38393d9b4 : SUCCESS in 42m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b64c6b745033460da423804c75627fd6 : SUCCESS in 55m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"0896f6340ddbc44454b21c921378f3621527787b","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 05:56:28.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-8-fc : FAILURE in 6m\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-01-995001-8-iscsi : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":8},{"id":"465c6251e7aacc02759d75a78f0b221c3c862a41","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-09-10 23:46:10.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8aca00bf61cb4ccc455e61e27d572b9aaa52add8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 00:14:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2cee4d497d746e091d65389f464b393\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2112fa2535bd4e64a6c77b195fb193e5 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd278dc825834992b6e6568eca0982dc : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bbf01b7045f04afbb9e512459eb2b923 : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3cd00e9a61a77142f1a4fd4379fe14d143990b9f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 00:21:53.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b8ef8955a6074d82bee5d9eaaedc5012/","accounts_in_message":[],"_revision_number":8},{"id":"b750b75044553b390109a1859b3455f5fb5519e3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-11 00:31:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b8ef8955a6074d82bee5d9eaaedc5012\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ee9624de7d84551a02951baf2d01013 : SUCCESS in 43m 56s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ebb7e218bfa4636a6854c1abaf82aa8 : SUCCESS in 35m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f39c00a606d49d490f30f723cf08aed : SUCCESS in 42m 34s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8cad68ec0ca646059f28505854eec89c : SUCCESS in 36m 02s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e3603625680b412abf4baa931d4c5dec : FAILURE in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e8d911026f173177568c169824413375ae02ab0c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-11 00:38:00.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/aaebc11374dc4701a780d61cdc66ae84 : SUCCESS in 46m 14s\n- cinder-storpool-tempest-iscsi-multipath build/c09dfce6d2dd4d7e87ae27e88bbab26a : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":8},{"id":"a04b8bb805c842da34c4045656152d0b0e74dc98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 01:45:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d394353342214df48a65e28ef424ee45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eab6576df3a4150a64fa28b7c18d8eb : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f7e5dff95d54d6981d7384a852f656e : SUCCESS in 9m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73d60d8679a7437b9d5c391e751d4afb : SUCCESS in 10m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1d59a60c2c94acd908ef3a9f475e5f5 : TIMED_OUT in 40m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a95b82a0254333967e0d476cebbf68 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/01e93824aa694e1e94a117d8b6bb7dd3 : SUCCESS in 1h 02m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ac9910cb21f46e3b3fad654eea1b734 : SUCCESS in 1h 06m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ce545ac5c3d43a6a2cf32acf5a43622 : SUCCESS in 1h 22m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c49a6bb9a1224b21b4d1ad57568fec7a : SUCCESS in 1h 01m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d08864b60df94a87a3d82d8904f47e05 : SUCCESS in 10m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59f4fa823e604ecba9bc7adaf3f009f2 : TIMED_OUT in 41m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a77bbbc53170496b9a75ab00ebb7c7e4 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c829985950fb4aeaac1cade2d6487054 : SUCCESS in 3m 26s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7f762295631c4597bf2bb298ce627a6d : SUCCESS in 12m 45s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/798e9f49b28649c9ab434271a4e0b914 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ca64f1d49484447a3e4a32a72f618aa : SUCCESS in 9m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e3ccbd1317bd40a8b8a27776670eb475 : SUCCESS in 5m 58s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/526068eb4add464cb938ac50285794ee : FAILURE in 5m 35s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6199462f81b0420ea66590afc4eee462 : SUCCESS in 8m 49s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e302e2f7e13141028bae205a391198f5 : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04d0cd32e40b479cbb77c155faf4b712 : FAILURE in 9m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ab24d35d38e4255bcadcbca8102cdb5 : SUCCESS in 1h 27m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2344eb7633b342c591176698116ca3fa : SUCCESS in 1h 09m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/abbb15bad0b248f68808a0e7e3f7263c : SUCCESS in 1h 52m 52s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/90aa7be187ca4a5da4d9d045dfdc2cbf : SUCCESS in 1h 18m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/01073c4e9a57462a84cd1c39f72eb1ba : SUCCESS in 1h 39m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/8e5edb9fa0674acab0902599a5c53521 : SUCCESS in 34m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bd709816a2c24c35a2cc7313c2df91b8 : SUCCESS in 1h 19m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3b73b5eafaa7485d86f99f13fae53987 : SUCCESS in 1h 15m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/204489a8695f4ff09e776c3b4fc9a877 : SUCCESS in 31m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f647f47019084acba6ed9efe3d73f924 : SUCCESS in 1h 24m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/915ae23c66464135aec8db52e57cab3a : SUCCESS in 1h 28m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6e1600998144daf941903094c2af30a : SUCCESS in 56m 34s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"d4893f05e5602ae5af73778681aa9441718fddef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-11 01:55:05.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/11_09_2026_01_48_56\nLeave a comment with","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"54c7a895095062449c90177bd77a6a1a7e6a951f","revisions":{"aab6e1823d05298c76648b8bb4fabe90aacfe426":{"kind":"REWORK","_number":1,"created":"2026-06-26 05:36:38.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/1"}}},"commit":{"parents":[{"commit":"34c049bccdaf1aeed49d53207730b96e6ed5ff11","subject":"WIP: Run Cinder\u0027s Backup service in native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/34c049bccdaf1aeed49d53207730b96e6ed5ff11"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-06-26 05:36:24.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-06-26 05:36:24.000000000","tz":-240},"subject":"image_utils: replace direct tpool usage with concurrency helpers","message":"image_utils: replace direct tpool usage with concurrency helpers\n\nReplace unconditional eventlet tpool usage in image_utils.py with\nthe concurrency-mode-aware helpers from cinder.utils, enabling the\nmodule to run under both eventlet and native threading.\n\n- Remove top-level \u0027from eventlet import tpool\u0027 import\n- Guard tpool.execute() in verify_glance_image_signature() on\n  concurrency_mode_threading(): call _verify_image directly when\n  running native threads, use tpool.execute with a deferred import\n  when running under eventlet\n- Replace four tpool.Proxy() wraps in fetch() and upload_volume()\n  with utils.tpool_wrap(), which returns the object unchanged in\n  threading mode and wraps it with tpool.Proxy under eventlet\n\nGenerated-By: claude-opus-4-6 (OpenCode)\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aab6e1823d05298c76648b8bb4fabe90aacfe426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aab6e1823d05298c76648b8bb4fabe90aacfe426"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"34c049bccdaf1aeed49d53207730b96e6ed5ff11","is_merged_in_target_branch":false,"change_id":"I6320b495e7da077fb7204b50c8d06d736a36e946","change_number":977851,"patch_set_number":18,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5bcf6367e54534f7c009e45b1dd363813d773cc4":{"kind":"REWORK","_number":2,"created":"2026-07-10 03:12:53.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/2"}}},"commit":{"parents":[{"commit":"38a4b9877b2cb14a8b051e0d647a70dce046c462","subject":"Use futurist-based thread pool for backup manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38a4b9877b2cb14a8b051e0d647a70dce046c462"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"subject":"image_utils: remove eventlet tpool usage entirely","message":"image_utils: remove eventlet tpool usage entirely\n\nRemove all eventlet tpool usage from image_utils.py. Since cinder is\nremoving eventlet entirely, there is no need for conditional fallbacks\nor tpool_wrap helpers. Call functions directly and pass file objects\nwithout tpool.Proxy wrapping.\n\n- Remove \u0027from eventlet import tpool\u0027 import\n- Replace tpool.execute(_verify_image, ...) with direct call\n- Replace tpool.Proxy(file_obj) with bare file_obj (4 locations)\n- Update tests: remove tpool.Proxy/tpool.execute mocks, assert on\n  direct file objects instead of mock_proxy.return_value\n\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\nSigned-off-by: Adam Harwell \u003cadamharwell@geico.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bcf6367e54534f7c009e45b1dd363813d773cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5bcf6367e54534f7c009e45b1dd363813d773cc4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38a4b9877b2cb14a8b051e0d647a70dce046c462","is_merged_in_target_branch":false,"change_id":"Iec74fc309936b8b4714274fbc2883e3c616aaf6e","change_number":995140,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"72bb194bf1ade87a9b73ec2d137f4578f38da7c8":{"kind":"REWORK","_number":3,"created":"2026-07-11 01:19:56.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/3"}}},"commit":{"parents":[{"commit":"38a4b9877b2cb14a8b051e0d647a70dce046c462","subject":"Use futurist-based thread pool for backup manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38a4b9877b2cb14a8b051e0d647a70dce046c462"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-11 01:19:28.000000000","tz":-240},"subject":"Remove eventlet tpool usage from image_utils and chunkeddriver","message":"Remove eventlet tpool usage from image_utils and chunkeddriver\n\nReplace direct eventlet.tpool.execute() and eventlet.sleep(0) calls\nwith threading-safe equivalents:\n\n- image_utils.py: remove tpool.execute wrapping of _verify_image and\n  tpool.Proxy wrapping of file objects (already on a worker thread\n  under native threading; tpool is unnecessary)\n- chunkeddriver.py: replace eventlet.tpool.execute with\n  utils.tpool_wrap (returns object unchanged under native threading,\n  wraps with tpool.Proxy under eventlet) and eventlet.sleep(0) with\n  time.sleep(0)\n\nUnder native threading (OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue),\nthe raw eventlet calls create orphan eventlet hubs on futurist\nThreadPoolExecutor worker threads.  When multiple backup operations\nrun concurrently, these hubs conflict with each other, producing\n\u0027greenlet.error: Cannot switch to a different thread\u0027 and permanently\nwedging backup operations.\n\nGenerated-By: OpenCode (claude-opus-4.6)\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72bb194bf1ade87a9b73ec2d137f4578f38da7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72bb194bf1ade87a9b73ec2d137f4578f38da7c8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38a4b9877b2cb14a8b051e0d647a70dce046c462","is_merged_in_target_branch":false,"change_id":"Iec74fc309936b8b4714274fbc2883e3c616aaf6e","change_number":995140,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"6a628fb512280bec632349227b2d9b1ff94a3104":{"kind":"REWORK","_number":4,"created":"2026-07-11 08:09:25.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/4"}}},"commit":{"parents":[{"commit":"38a4b9877b2cb14a8b051e0d647a70dce046c462","subject":"Use futurist-based thread pool for backup manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38a4b9877b2cb14a8b051e0d647a70dce046c462"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-11 08:09:15.000000000","tz":-240},"subject":"Remove eventlet tpool usage from image_utils, chunkeddriver, and manager","message":"Remove eventlet tpool usage from image_utils, chunkeddriver, and manager\n\nReplace direct eventlet.tpool.execute(), eventlet.tpool.Proxy(), and\neventlet.sleep(0) calls with threading-safe equivalents:\n\n- image_utils.py: remove tpool.execute wrapping of _verify_image and\n  tpool.Proxy wrapping of file objects (already on a worker thread\n  under native threading; tpool is unnecessary)\n- chunkeddriver.py: replace eventlet.tpool.Proxy with utils.tpool_wrap\n  (returns object unchanged under native threading, wraps with\n  tpool.Proxy under eventlet) and eventlet.sleep(0) with time.sleep(0)\n- backup/manager.py: replace tpool.Proxy(device_file/device_path) with\n  utils.tpool_wrap (same conditional pattern); remove \u0027from eventlet\n  import tpool\u0027\n\nUnder native threading (OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue),\nthe raw eventlet calls create orphan eventlet hubs on futurist\nThreadPoolExecutor worker threads.  When multiple backup operations\nrun concurrently, these hubs conflict with each other, producing\n\u0027greenlet.error: Cannot switch to a different thread\u0027 and permanently\nwedging backup operations.\n\nGenerated-By: OpenCode (claude-opus-4.6)\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a628fb512280bec632349227b2d9b1ff94a3104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6a628fb512280bec632349227b2d9b1ff94a3104"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38a4b9877b2cb14a8b051e0d647a70dce046c462","is_merged_in_target_branch":false,"change_id":"Iec74fc309936b8b4714274fbc2883e3c616aaf6e","change_number":995140,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"214cfa2579b7d7ad3143723cd9b5920335b8d19b":{"kind":"REWORK","_number":5,"created":"2026-07-24 04:32:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/5"}}},"commit":{"parents":[{"commit":"38a4b9877b2cb14a8b051e0d647a70dce046c462","subject":"Use futurist-based thread pool for backup manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/38a4b9877b2cb14a8b051e0d647a70dce046c462"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-24 04:32:14.000000000","tz":-240},"subject":"Remove eventlet tpool usage from image_utils, chunkeddriver, and manager","message":"Remove eventlet tpool usage from image_utils, chunkeddriver, and manager\n\nReplace direct eventlet.tpool.execute(), eventlet.tpool.Proxy(), and\neventlet.sleep(0) calls with threading-safe equivalents:\n\n- image_utils.py: remove tpool.execute wrapping of _verify_image and\n  tpool.Proxy wrapping of file objects (already on a worker thread\n  under native threading; tpool is unnecessary)\n- chunkeddriver.py: replace eventlet.tpool.Proxy with utils.tpool_wrap\n  (returns object unchanged under native threading, wraps with\n  tpool.Proxy under eventlet) and eventlet.sleep(0) with time.sleep(0)\n- backup/manager.py: replace tpool.Proxy(device_file/device_path) with\n  utils.tpool_wrap (same conditional pattern); remove \u0027from eventlet\n  import tpool\u0027\n\nUnder native threading (OS_CINDER_DISABLE_EVENTLET_PATCHING\u003dtrue),\nthe raw eventlet calls create orphan eventlet hubs on futurist\nThreadPoolExecutor worker threads.  When multiple backup operations\nrun concurrently, these hubs conflict with each other, producing\n\u0027greenlet.error: Cannot switch to a different thread\u0027 and permanently\nwedging backup operations.\n\nGenerated-By: OpenCode (claude-opus-4.6)\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/214cfa2579b7d7ad3143723cd9b5920335b8d19b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/214cfa2579b7d7ad3143723cd9b5920335b8d19b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"38a4b9877b2cb14a8b051e0d647a70dce046c462","is_merged_in_target_branch":false,"change_id":"Iec74fc309936b8b4714274fbc2883e3c616aaf6e","change_number":995140,"patch_set_number":2,"change_status":"ABANDONED"}],"branch":"refs/heads/master"},"ce792840e84c832ce6b5b356de7cb85d26fac263":{"kind":"REWORK","_number":6,"created":"2026-08-12 06:01:41.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/6"}}},"commit":{"parents":[{"commit":"c07c49c586d1667d3e11a7142689894fd9196c74","subject":"Run Cinder\u0027s Backup service in native threads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c07c49c586d1667d3e11a7142689894fd9196c74"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-08-12 05:07:59.000000000","tz":-240},"subject":"image_utils: use concurrency-aware tpool wrapping","message":"image_utils: use concurrency-aware tpool wrapping\n\nReplace direct eventlet tpool use with utils.tpool_wrap. The helper\nretains tpool behavior when running with eventlet and returns callable\nand file objects unchanged with native threads.\n\nAdd coverage for the helper behavior in both concurrency modes and\nupdate image_utils tests to verify the wrapped objects passed to Glance.\n\nGenerated-By: OpenCode claude-opus-4.6\nAssisted-By: Codex GPT-5\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce792840e84c832ce6b5b356de7cb85d26fac263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce792840e84c832ce6b5b356de7cb85d26fac263"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c07c49c586d1667d3e11a7142689894fd9196c74","is_merged_in_target_branch":true,"change_id":"I6320b495e7da077fb7204b50c8d06d736a36e946","change_number":977851,"patch_set_number":25,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4dd78b653dba251ce2faa5ce239504f784d7176a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-03 20:21:08.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/01/995001/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/7"}}},"commit":{"parents":[{"commit":"5dee715ca2f26fd768fe58bf3f727e78e5587567","subject":"Merge \"Fix domain ids lost during rpc call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5dee715ca2f26fd768fe58bf3f727e78e5587567"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-09-03 20:21:08.000000000","tz":0},"subject":"image_utils: use concurrency-aware tpool wrapping","message":"image_utils: use concurrency-aware tpool wrapping\n\nReplace direct eventlet tpool use with utils.tpool_wrap. The helper\nretains tpool behavior when running with eventlet and returns callable\nand file objects unchanged with native threads.\n\nAdd coverage for the helper behavior in both concurrency modes and\nupdate image_utils tests to verify the wrapped objects passed to Glance.\n\nGenerated-By: OpenCode claude-opus-4.6\nAssisted-By: Codex GPT-5\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4dd78b653dba251ce2faa5ce239504f784d7176a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4dd78b653dba251ce2faa5ce239504f784d7176a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5dee715ca2f26fd768fe58bf3f727e78e5587567","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c07c49c586d1667d3e11a7142689894fd9196c74","ours":"ce792840e84c832ce6b5b356de7cb85d26fac263","theirs":"5dee715ca2f26fd768fe58bf3f727e78e5587567","merge_strategy":"recursive","contains_conflicts":false}},"54c7a895095062449c90177bd77a6a1a7e6a951f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-09-09 19:08:00.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"real_uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/01/995001/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/01/995001/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/01/995001/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/01/995001/8"}}},"commit":{"parents":[{"commit":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","subject":"Merge \"[SVf]: Enable support to extend a volume with snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/31ec30d44f086bcd2e60fad0c4cbc461a1044518"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-10 03:12:48.000000000","tz":-240},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-09-09 19:08:00.000000000","tz":0},"subject":"image_utils: use concurrency-aware tpool wrapping","message":"image_utils: use concurrency-aware tpool wrapping\n\nReplace direct eventlet tpool use with utils.tpool_wrap. The helper\nretains tpool behavior when running with eventlet and returns callable\nand file objects unchanged with native threads.\n\nAdd coverage for the helper behavior in both concurrency modes and\nupdate image_utils tests to verify the wrapped objects passed to Glance.\n\nGenerated-By: OpenCode claude-opus-4.6\nAssisted-By: Codex GPT-5\nSigned-off-by: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: Ia13b86fb5a16af6bfe1e868f645dbf240ed92832\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54c7a895095062449c90177bd77a6a1a7e6a951f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/54c7a895095062449c90177bd77a6a1a7e6a951f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5dee715ca2f26fd768fe58bf3f727e78e5587567","ours":"4dd78b653dba251ce2faa5ce239504f784d7176a","theirs":"31ec30d44f086bcd2e60fad0c4cbc461a1044518","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
