)]}'
{"id":"openstack%2Fcinder~996722","triplet_id":"openstack%2Fcinder~master~I09503b7dd1a5897da56e6b7127eefebf76b2459f","project":"openstack/cinder","branch":"master","topic":"driver-security","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-08-22 21:30:42.000000000","reason":"Change was submitted"},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-08-13 15:40:43.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"I09503b7dd1a5897da56e6b7127eefebf76b2459f","subject":"[docs] Add security issues to driver compliance","status":"MERGED","created":"2026-07-09 21:57:13.000000000","updated":"2026-08-22 23:27:11.000000000","submitted":"2026-08-22 21:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996722-driver-security","meta_rev_id":"a3d46ce17ea953f67aa0add35850b92a2b5c1cf1","_number":996722,"virtual_id_number":996722,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-22 21:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2026-08-12 14:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2026-08-19 21:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-08-22 20:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-08-13 15:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":1,"date":"2026-08-18 09:16:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":1,"date":"2026-08-22 20:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-09 22:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 22:19:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-09 22:58:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-09 23:24:45.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-10 06:46:10.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-10 09:15:08.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-08-12 14:32:38.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-08-12 14:54:00.000000000","updated_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-08-18 09:16:42.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"REVIEWER"},{"updated":"2026-08-19 21:15:16.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-08-22 20:18:57.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"e67e0e6427806dd7d5303c607ede8dd984dc2e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-09 21:57:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de96f1cdf6fe7c159469a02acd3555520d96a70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 22:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd575278c8804de3b5004ec0b443c0b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834d5d5266db43afbc50f02a5ad285f1 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205ce4f1f684415cb8d2f73bf0e04bef : SUCCESS in 15m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc78f7fe0dd94cc8aa8dc704a5697cbe : SUCCESS in 8m 02s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dd6e9edc783441c0b9daab7803fdf7e3 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d89bca7202bf193a8bf0437ae9a7994f159e39c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-09 22:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_07_2026_22_22_38\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"cfb9252c8f4b746e747890c5726ccfa040626daa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-09 22:58:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I09503b7dd1a5897da56e6b7127eefebf76b2459f .","accounts_in_message":[],"_revision_number":1},{"id":"06622b4c3b53636137e9ec08122b974bb4e1d28d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 23:24:45.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/833 : SUCCESS in 0h 42m 39s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/996722","accounts_in_message":[],"_revision_number":1},{"id":"3c109e95408950b07be4ca46af949126bce98208","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 23:26:23.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/563 : SUCCESS in 0h 43m 37s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/996722","accounts_in_message":[],"_revision_number":1},{"id":"ce8fe0d336a46e9adea5bc39e3a91d227cd63c85","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-09 23:26:34.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/477 : SUCCESS in 0h 43m 21s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/996722","accounts_in_message":[],"_revision_number":1},{"id":"fab4cb43e13d360d4dddcf09291f2e4db5993b79","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 06:46:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4755/.","accounts_in_message":[],"_revision_number":1},{"id":"9c12fb272958caa40c2c83fd6581c6914ad1251e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-10 09:15:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-996722-1-fc : SUCCESS in 1h 5m 26s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-996722-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"afc95ba2d828b2ab31b73ccd6d129b4ae4b90df0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-10 11:19:05.000000000","message":"Topic set to driver-security","accounts_in_message":[],"_revision_number":1},{"id":"08c4816bed37fbcea0be61d750abc9f001ad1859","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 21:25:19.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4579/.","accounts_in_message":[],"_revision_number":1},{"id":"2691c195e729f546387fcafdbf4c961dce565bc3","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-08-12 14:32:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"233e55b85c10d97d6b2fc210e0babe5d15acfbe1","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-13 15:40:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76427903bde84cf3f2e533bc953a720479baf38c","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2026-08-18 09:16:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3580ef6d43b8ca703918eb528bb14db846fb6db0","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-19 21:15:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75c4d75bc949ddecad0b83d90635a32b3630e479","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-22 20:18:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd3445964374fabf02a58ffeb9b49a0635e8593c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 20:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfc8e3a9d3a0eeaf226b9eddbf525fc1e25dc175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 21:30:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e877a7afc31f486e83cefa963ad6b735\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b6542621038487c95a4fb22fc15fdab : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd21850f5ff545d7981f780cf8248e30 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c415f760f1433a9af2c03463f3a96c : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5c9839a5add8bd604e94d3b2dd0513f18915e25a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 21:30:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3d46ce17ea953f67aa0add35850b92a2b5c1cf1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-22 23:27:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-996722-1-fc : FAILURE in 11m 16s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-996722-1-iscsi : SUCCESS in 1h 10m 6s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d0449970716ebd36e28d34caf3c58491fadcc2a","revisions":{"9d0449970716ebd36e28d34caf3c58491fadcc2a":{"kind":"REWORK","_number":1,"created":"2026-07-09 21:57:13.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/22/996722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/996722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/996722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/996722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/996722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/996722/1"}}},"commit":{"parents":[{"commit":"fdccb4f23a4c51990fb1c5192120671bf45cbabc","subject":"[docs] Eventlet is not a preferred threading model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fdccb4f23a4c51990fb1c5192120671bf45cbabc"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-07-09 21:39:37.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-07-09 21:39:37.000000000","tz":-240},"subject":"[docs] Add security issues to driver compliance","message":"[docs] Add security issues to driver compliance\n\nUpdate the driver compliance policy to include security issues.\nThis will allow us to handle security compliance in the same way\nas third-party CI compliance, that is, by setting SUPPORTED\u003dFalse\nin the driver code.  Because operators will have to set\n\u0027enable_unsupported_driver\u003dTrue\u0027 in the driver\u0027s section of the\ncinder configuration file in order to use the driver, operators\nwill be making a conscious choice to use an out-of-compliance\ndriver.\n\nThis policy is stated to go into effect beginning with the Indri\nrelease, which should allow the maintainers of current cinder\ndrivers to make any appropriate fixes to make sure their drivers\nare in security compliance.\n\nChange-Id: I09503b7dd1a5897da56e6b7127eefebf76b2459f\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d0449970716ebd36e28d34caf3c58491fadcc2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d0449970716ebd36e28d34caf3c58491fadcc2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
