)]}'
{"id":"openstack%2Fcinder~996837","triplet_id":"openstack%2Fcinder~master~I5a3ff2051c61acbf6e0e627df841327b219e8f74","project":"openstack/cinder","branch":"master","topic":"fix/empty-img-cache-size","attention_set":{"37477":{"account":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"last_update":"2026-07-10 17:48:51.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5a3ff2051c61acbf6e0e627df841327b219e8f74","subject":"fix: img cleanup job stuck on empty img size","status":"NEW","created":"2026-07-10 17:02:30.000000000","updated":"2026-07-11 10:32:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf0c98b27b3bdaf1da4c54096422d737883dbb8f","_number":996837,"virtual_id_number":996837,"owner":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-10 19:51:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 17:48:51.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-10 18:01:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-10 18:04:42.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-10 19:51:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-10 20:03:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-10 20:27:34.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-10 23:10:38.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-11 10:32:03.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"}],"messages":[{"id":"682b81d177328668118fad2c3baef87fab53f6b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"date":"2026-07-10 17:02:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec97313d6f6c375af4a49238c435f9811056925d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 17:48:51.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b739b06300ad434b9a098d51074c64d0/","accounts_in_message":[],"_revision_number":1},{"id":"39d3502fc39df2799af6bfffed17df237c13f5f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-10 17:53:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b739b06300ad434b9a098d51074c64d0\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/127210dd265c4dcc82aa87c679b65aed : SUCCESS in 49m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f00282e60b944946bb1f63259da2ee3f : SUCCESS in 42m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7eef753797eb4de2addecc3bf97f889a : SUCCESS in 39m 23s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f63f9d16f036470491d79c7eb40941dc : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d7ec29908f106f866662c99e40486db7b8e534fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-10 18:01:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I5a3ff2051c61acbf6e0e627df841327b219e8f74 .","accounts_in_message":[],"_revision_number":1},{"id":"3e20295e02baafa5ecbb234cc95273a59824a64f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 18:04:42.000000000","message":"Patch Set 1:\n\nHPE AlletraMP NVMe/TCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/491 : SUCCESS in 0h 41m 40s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/996837","accounts_in_message":[],"_revision_number":1},{"id":"ffd2b753f883551e1a4643eb656112973b997450","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 18:05:19.000000000","message":"Patch Set 1:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/577 : SUCCESS in 0h 42m 3s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/996837","accounts_in_message":[],"_revision_number":1},{"id":"c607599e087cdfb3b526d34264422b0dc7545869","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"HPE AlletraMP iSCSI CI","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-10 18:06:30.000000000","message":"Patch Set 1:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/847 : SUCCESS in 0h 43m 17s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/996837","accounts_in_message":[],"_revision_number":1},{"id":"a4a6930d747730aaea498464233fa6b214dd7a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 19:51:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c18b423d4fe488b98fcb4efdd007699\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ba0df39a4d4dbd9a045d0715b67923 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95d6a2fd058e46f191f8133a42b5987f : SUCCESS in 9m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6044df03d7e94f3a84af1498ea50e7e5 : SUCCESS in 7m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93843f5f20fa45c0b8db165240fd7e85 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831fcb22ae204e27a2b69ad2c552cc66 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3cebbf70db054bc6a2ee38bef1a961e2 : SUCCESS in 1h 05m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82b5d42818434de091498ed4ea216917 : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2a480ae8c514f13bab828f4f4cc2c2e : SUCCESS in 1h 51m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fad98d66cc442888fae8aa393f765a3 : SUCCESS in 1h 05m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a958a8de097431181160b8ed998e5a8 : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/79bbaf86888847e2b8ec5d0598cf3cec : SUCCESS in 21m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/567a90ec5f69415480db3e9148bdbfc9 : SUCCESS in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1b25906e442c492e8aec718f5fc9a626 : SUCCESS in 6m 24s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/f4305effe07645e0a27a7a16e7091a83 : FAILURE in 12m 42s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/480bf1950a8a468a9a83ba47f9383417 : SUCCESS in 6m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/97ce8110375a4ca0b8f3a1f52dd286e3 : SUCCESS in 6m 02s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/87c20c2bb4ab46aab0e0e9472d017662 : SUCCESS in 8m 26s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/71473b0f513b4e05bceea3c492eb8502 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/329bc29ca824452a99795ab76ed51d18 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc415ca225fd4d67b588e7d29e8fd2e6 : SUCCESS in 1h 02m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/daf376fcf0bd46ec95288028263e35e2 : FAILURE in 1h 19m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d49da5d8ffb9464ebf51d7540a14e0e3 : SUCCESS in 1h 54m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/347da89ea147462694c48a377a256bfc : FAILURE in 2h 43m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a51388f6292642d4a3b5b5fe521a1833 : SUCCESS in 2h 35m 00s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/54620579e40b430d9c436a48ec184070 : SUCCESS in 19m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dec9895b59984463abf15113e8e2099d : SUCCESS in 1h 17m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d7282383747f41b0ac15c55a329cd52e : SUCCESS in 1h 50m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/38c158bbdcce4c80a2683b402c80d4a6 : SUCCESS in 32m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9099a2887bcf4942a6bafb7b8118df13 : SUCCESS in 1h 39m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20496b12f8b94685b2b64c2dfacd90c2 : SUCCESS in 59m 47s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"f50dccbfdb1e71d3ae3556a21ac7fefc8d57c0f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-10 20:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_07_2026_20_07_52\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"40fd3283584776a7e3e7135ca48f2e3551a09a22","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-10 20:27:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/48454bba6ccb4f96a321f01b680d6c5e : SUCCESS in 56m 24s\n- cinder-storpool-tempest-iscsi-multipath build/49c814d0673c421ba1ec391d55b3e274 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0f2a704fd1a0445c29aaae187c4bf02ea5a2fcce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 22:23:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dde04b704494aa0b55d2ee5781613b0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca5fe31365124185b275e0bdb4a26fa2 : SUCCESS in 20m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0c96a5de46b54ef79b1c717a2af98c99 : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1520430af5084b699a89637032e27fc8 : TIMED_OUT in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f205c0dbd8e076aed86d06b8f1cf079de5081900","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-10 23:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4769/.","accounts_in_message":[],"_revision_number":1},{"id":"bf0c98b27b3bdaf1da4c54096422d737883dbb8f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-11 10:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-996837-1-fc : SUCCESS in 1h 27m 36s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-37-996837-1-iscsi : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a93c9d3db0e1fc21312825e849762c963e60f42b","revisions":{"a93c9d3db0e1fc21312825e849762c963e60f42b":{"kind":"REWORK","_number":1,"created":"2026-07-10 17:02:30.000000000","uploader":{"_account_id":37477,"name":"Florian","display_name":"FloD","email":"Florian.Doerheim@digits.schwarz","username":"FloD"},"ref":"refs/changes/37/996837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/37/996837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/996837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/996837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/37/996837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/37/996837/1"}}},"commit":{"parents":[{"commit":"fb302038e5ceaa5271fd1f240d9aae0243b5a92d","subject":"Merge \"Bump mypy to 2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fb302038e5ceaa5271fd1f240d9aae0243b5a92d"}]}],"author":{"name":"Florian Dörheim","email":"Florian.Doerheim@digits.schwarz","date":"2026-07-10 15:08:03.000000000","tz":120},"committer":{"name":"Florian Dörheim","email":"Florian.Doerheim@digits.schwarz","date":"2026-07-10 17:00:28.000000000","tz":120},"subject":"fix: img cleanup job stuck on empty img size","message":"fix: img cleanup job stuck on empty img size\n\nwhen cleaning up the img-cache files on the cinder-side shares, there is\na possibility of fetching a file that has already been deleted on the\nnetapp side. This causes the received u_bytes value to be None, which\nwill crash the cleanup job upon sorting the list for the biggest files\n\nChange-Id: I5a3ff2051c61acbf6e0e627df841327b219e8f74\nSigned-off-by: Florian Dörheim \u003cFlorian.Doerheim@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a93c9d3db0e1fc21312825e849762c963e60f42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a93c9d3db0e1fc21312825e849762c963e60f42b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
