)]}'
{"id":"openstack%2Fcinder~997422","triplet_id":"openstack%2Fcinder~master~I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2","project":"openstack/cinder","branch":"master","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-10-02 14:55:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2026-07-16 13:44:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2026-07-16 13:44:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-10-05 04:29:04.000000000","reason":"Someone else replied on the change"},"38623":{"account":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"last_update":"2026-07-16 13:44:15.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-10-02 14:10:41.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"38059":{"account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"last_update":"2026-09-24 05:22:16.000000000","reason":"\u003cGERRIT_ACCOUNT_38059\u003e replied on the change","reason_account":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}}},"hashtags":[],"change_id":"I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2","subject":"NetApp ONTAP: Add REST metrics-based pool utilization","status":"NEW","created":"2026-07-15 16:15:30.000000000","updated":"2026-10-05 06:48:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":58,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc7af6e64f45f88ebce67d8bf4e91ab7dd892d9b","_number":997422,"virtual_id_number":997422,"owner":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-05 06:38:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":1,"date":"2026-10-02 14:26:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":2,"date":"2026-10-02 15:17:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":1,"date":"2026-09-24 05:22:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-15 16:53:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-15 17:13:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-15 17:56:29.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-15 18:02:44.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-15 18:39:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-15 18:43:29.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-15 18:47:11.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-15 18:52:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-15 20:13:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-15 20:14:06.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-16 07:26:04.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-16 08:15:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-16 13:44:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38623,"name":"Sanketh Shetty","display_name":"Sanketh Shetty","email":"sanketh.shetty@netapp.com","username":"sanketh-netapp","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-16 13:44:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2026-07-16 13:44:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-16 13:44:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"state":"REVIEWER"},{"updated":"2026-07-16 13:44:15.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-07-17 15:56:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-01 09:36:37.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-02 06:09:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-09-22 04:44:27.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"8eb7dc498b77ffe3e665d1f564e2db6d1dc8e1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 16:15:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c38d46448024010fe53a24de06048d54c258f7fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-15 16:53:20.000000000","message":"Patch Set 1:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/74e94239dc5a4ddd996867b3eae9de91/","accounts_in_message":[],"_revision_number":1},{"id":"cf122fe37ee0845873b55b1b665796f1be203b39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-15 17:10:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75dfd1fb2488c68c3090a6a79d12dce7e47694a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-15 17:13:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":1},{"id":"1f7375fdcb831ec9fdd1d86a7d4f5be22ef3aa3f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-15 17:42:42.000000000","message":"Patch Set 2:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/1e5be65aebef499e86fc29934a6e7a67/","accounts_in_message":[],"_revision_number":2},{"id":"7a700b6556a5dc147b1cf12495cec5f1a4cfa763","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-15 17:52:43.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1e5be65aebef499e86fc29934a6e7a67\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1a96edf2bf344e748c61cf0dce74bde8 : FAILURE in 41m 59s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/da2defe5a4334f469141a69d80a17e1e : SUCCESS in 33m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7674139015ad4f3192b3def4374b6995 : FAILURE in 33m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d3fa72c5c780429e9b1527e909f74e30 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":2},{"id":"16cb3a3d621f0b630eac04b5dbb8ccd200cba99c","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-15 17:56:29.000000000","message":"Patch Set 2:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/529 : FAILURE in 0h 45m 30s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":2},{"id":"5e8c3a1ed8cb83afceba472cd2fd0433cd414402","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-15 18:02:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest build/61df98d3d5ab4ef1b4d58f4c482907ce : FAILURE in 45m 54s\n- cinder-storpool-tempest-iscsi-multipath build/5fda1fc7347046f791aa91416d08585c : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":2},{"id":"dd93e2b9ac4ab323ea205ea1bacf61e58dd4300b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-15 18:04:58.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":2},{"id":"906ef98ffd8dd600ae92c36fac59db0a90b82f73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 18:39:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d088af34cfcb463bbddac72c9b0d81df\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/250a4ee288894d4ca5ceb2b5e93dc06c : SUCCESS in 18m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b90b4395c8284744ba48f65f586e8a69 : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/09e983a061ee46198855bcbe6911943d : SUCCESS in 13m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b624105aad5eb0f0dcd79d521f4f98c063a4c545","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-15 18:43:29.000000000","message":"Patch Set 2:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/884 : SUCCESS in 0h 44m 27s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":2},{"id":"55f7f5a81f3cd315347040d44702f31636da064d","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-15 18:47:11.000000000","message":"Patch Set 2:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/621 : SUCCESS in 0h 45m 33s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":2},{"id":"b3bfdc817a8f082351984ea58faa9cf2fe2bac0f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-15 18:52:23.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4804/.","accounts_in_message":[],"_revision_number":1},{"id":"3b7a83edfe124016bc5773b7c4968eecac682803","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-15 19:02:02.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4805/.","accounts_in_message":[],"_revision_number":2},{"id":"d6a43a0bb671fb4c2410d2a3c8dc5bba0ed48a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 20:13:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d44c3e22a684a6ab0e72fa9ca2a13fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdcd91420d94128af78774a694ac5fd : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6175953e711d4e2b8be840d91e745a81 : SUCCESS in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6846b480821a4239950eaf04ac0375a1 : SUCCESS in 7m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da860ea5665b48e6a32c28178b3a7b06 : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b291b06fd9cf45a1bce2172c4acdaf32 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/83a7dc59ca7e4502a56c172fb68acc8d : SUCCESS in 1h 01m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3297f6f3d47b461bba7174873b06bd52 : SUCCESS in 35m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b45998c6c89442f588818f7b00a89596 : SUCCESS in 1h 06m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04c8fdb4b48148f9bcc66ee6562f7c24 : SUCCESS in 32m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e7b6abdd1554421a88bfd6720849cfc : SUCCESS in 7m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b6ede3e78ec14bb6aded889813a1f478 : SUCCESS in 24m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9e314abe3eda486084b24bed7fcc8ece : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d6dfe178be34c96979b14b4e21aa925 : SUCCESS in 5m 51s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/467365d2306d4881b558a0477e1e18e5 : FAILURE in 9m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb3c8fa03bc74a7fa00364e4f03feb3b : SUCCESS in 8m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/adc60262e8614aa2ac29d081ffa647ee : SUCCESS in 6m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/4a62bb8db9b84a4294a8eedc415ee50f : SUCCESS in 10m 05s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/507f9eecb0b94492b38122a6d20b72ea : FAILURE in 52m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a043e118705a420abd1a24a9e4830b07 : SUCCESS in 4m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85c089743d8b48b9ac0102c1c0dcbd4d : SUCCESS in 58m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/20716514ebd34adfa1e96db31040df21 : FAILURE in 18m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf76deda50db4969b5b93c9e5bafa05e : SUCCESS in 1h 26m 29s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2a552538123c46dab943edf6ea91baf6 : FAILURE in 2h 58m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/a1a6236fe9ed43dab779ea32449cf380 : SUCCESS in 1h 19m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a08a4fe87ea54a5a8c91d6771e3ef61d : SUCCESS in 17m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ae84e0b2f18b4bf7957b5cc02597f23a : SUCCESS in 1h 22m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8947576e671d44bfa03b30144c95b8cc : SUCCESS in 1h 40m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2db56654c36f4fd79f75569f9a7abc31 : SUCCESS in 46m 53s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1af0f85c06e647b0853e5d02058663a3 : SUCCESS in 1h 31m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08e9b5090b2242d9b9c39950cb71d015 : SUCCESS in 29m 28s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"85c510cc7b2ebf13a90dec47ec0823fa9b40167f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-15 20:23:45.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_07_2026_20_17_24\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"6e3d7b6bc0937a42580299485fc95f321802339a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-16 07:26:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-2-fc : SUCCESS in 1h 4m 58s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"184deff0d441d1256bcb0b303d1302205e6fa5ee","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-16 08:15:43.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-1/FC/ : FAILURE in 15h 59m 25s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-1/iSCSI/ : SUCCESS in 3h 11m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"356253439c107abd1d5f941d6e20738cf3581940","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-16 08:55:47.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-2/FC/ : FAILURE in 15h 44m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-2/iSCSI/ : UNSTABLE in 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"04001aa5f1c457accf7241d0c6fcedb23f83c4c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-16 13:56:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bc82ba85d35bfa7102c5f09ddee4638f7be7f200","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 13:57:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3515f840c8cc4735a25c64be87ce3b54\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0db1b53bbbb9439689e2412812f3f0c2 : NODE_FAILURE Node(set) request 72fba9b9e0d04a6f8a2ea4f52dde0a39 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e8d75b8d75b74039bccc16f7caf9df0e : NODE_FAILURE Node(set) request 5cbbd7598aa5487cb804faee10c3dbec failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0b038258ee06429f8c77c19a6eb943b5 : NODE_FAILURE Node(set) request 156b54bc083a45d6b780fdc1b10b0f75 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d7c919e3855a717cd712556aef85104076a5d17","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 14:04:44.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4812/.","accounts_in_message":[],"_revision_number":3},{"id":"296cbcc2919cdb9ed8f12aa3b64a7a86c9c503f5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 14:33:38.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/22c5068d6fd44781a76d9df0cad236fb/","accounts_in_message":[],"_revision_number":3},{"id":"d9ce2a382b2a75458d0e8b42830b72b125716e75","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 14:50:05.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":3},{"id":"2c676a0105274a0980cb8b14d6332734b92d5160","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-16 14:54:36.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/889 : FAILURE in 0h 8m 34s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":3},{"id":"dab1c64ad3859f6f40158b8d37efab0295d130ae","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-16 15:00:39.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/626 : SUCCESS in 0h 45m 5s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":3},{"id":"19396899901383ccbc521420e7d1b37f5ef63024","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 15:11:21.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/22c5068d6fd44781a76d9df0cad236fb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ea5b9c3cd71140429f0a9a158a429fea : FAILURE in 35m 27s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa7a4daed3a848dcbd745520b9504365 : SUCCESS in 37m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e74d9e4ff30f49a3917bebc92c74b785 : SUCCESS in 34m 03s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/001097c1b29b49ab81b53ed17ce3382d : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":3},{"id":"718029e50cbaa1310650d22eebd88b9ec9aa737e","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-16 15:28:23.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/534 : SUCCESS in 0h 44m 41s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":3},{"id":"b86112f81d7681728306b3516d4cd0ea3f39e695","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-16 15:40:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/7c6eb7e0be054e91abf79db8dbd2f31b : SUCCESS in 41m 12s\n- cinder-storpool-tempest-iscsi-multipath build/067909fb94284bd09b239cbb2308bbbf : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":3},{"id":"dd5d340bd7e7cadc58fe69e8f3495b53b740286a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 17:11:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc47711c0d024224ac7ff0f101ae3b2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9283904c9c70478e8e674c419311f938 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/609993874fb3493998c79001bdf31796 : SUCCESS in 12m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95b5133757cd47b5a362e825c751f703 : SUCCESS in 10m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6ccd9d320d0845548a2491d1853cf678 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a7fa1f9aca4385a6395fa5246ec714 : SUCCESS in 15m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b2417a8f68dd4b52bf9cf525a062570b : SUCCESS in 40m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f000af10b773434dac4072bbe912cbe2 : SUCCESS in 1h 06m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18461bcd9a024870ab2b26e3e53d2797 : SUCCESS in 1h 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b627b1161949f2ad2e9c153d5ef0a2 : SUCCESS in 28m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/305ed39594c7465f9a001d6a983ca9d9 : SUCCESS in 10m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/292518b2b5a446c3a7ae4021ee1ead47 : FAILURE in 23m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c362b40d880e422e866f208591639dfc : SUCCESS in 3m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/92b2859f668044f8a93386c7bfb4e7a5 : SUCCESS in 7m 09s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/806e2ff6bf804a5087577c87e7da219e : FAILURE in 18m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f67952cdd1904deab782562355913ba4 : SUCCESS in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e7f574c2762446639e1d77402d7c60c7 : SUCCESS in 7m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7b4dc090d0274c0b9986f5724a1c3408 : SUCCESS in 6m 02s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1555a10d6e59471080a4c0e8c5f8e012 : FAILURE in 47m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/430d5204fda6488bb9b75c678961e1f6 : SUCCESS in 5m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68a54b9a7b434ae1ab30bbef6f7a04c3 : SUCCESS in 1h 36m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7ba6334f80a34d16bc5bd550ff8f7ad4 : FAILURE in 2h 28m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b18bbcdb31d949b981ca40eef1d12e20 : SUCCESS in 1h 22m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2cca805c3c5f4a5fa15715f0d433f182 : FAILURE in 3h 09m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/1da177a0973245589d57f0f8b1686c9f : SUCCESS in 1h 17m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/31e738cd3a6143d29a0b51c6c9a0f246 : SUCCESS in 20m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79522935aebd4fa39ff214715c910acc : SUCCESS in 1h 18m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86c18f30ab324dc2b37b100f36cf96b5 : SUCCESS in 1h 22m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/97fda870f7b447f6898e9f32f246f408 : SUCCESS in 45m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58e75ce488344b89ad05e1f0cac991b4 : SUCCESS in 1h 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d463477113148a8b6207a6a960afd96 : SUCCESS in 27m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"4ea60422027db7f0268f60b1bd3e967b31421624","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-16 17:21:19.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_07_2026_17_14_36\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"96d039d317d043e00db683d4c28bc4071c0bd47a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-17 06:06:45.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-3/FC/ : FAILURE in 16h 9m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-22-997422-3/iSCSI/ : SUCCESS in 2h 10m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"124df12a5bd68aae9a2806d8532ad864dc0d466b","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 14:57:34.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c3326fe758dc95cc83b13f52afea2bd36530a71b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-07-17 15:56:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3fa84db30e75d8b614c37b8e2e00fb1856b4f109","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 16:02:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb1463ca76826d942235f4128e34c0b23943ab7b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 07:14:28.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4628/.","accounts_in_message":[],"_revision_number":1},{"id":"b1ca2f3566620de0d887c6cf7c05e145aa49d975","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 08:49:05.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4636/.","accounts_in_message":[],"_revision_number":3},{"id":"822c2e81ecca24716a85a708d2185ec2a92213ba","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-19 14:29:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-3-fc : FAILURE in 11m 55s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-3-iscsi : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":3},{"id":"bb2cb9e00d7e5aafe04fb65bd264243fc6105c44","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-19 18:50:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-1-fc : FAILURE in 13m 20s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-1-iscsi : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a69c9b036814b3d3ae73381b74deef6046b1eef1","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-07-31 06:40:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cbd4cef776ca11bb6e71ec7e40fa8793bfc22eaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-13 18:32:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f2094f87daa81d4b5d97f995c990cad7f00e5701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 18:47:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae0b6e63a12c4ab4854546824bb61d4c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f11b3b963f44ac8becd131e722314aa : FAILURE in 9m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/717ea67c833f4b5194810dc616ec8ab3 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f409d74c583d4e55ace3e61a9e3e4b15 : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f21b20ee16ad0b4f8e853a4509252e6beda15413","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 19:11:06.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/4adf4a62ec064cb59cfe3f498126408f/","accounts_in_message":[],"_revision_number":4},{"id":"68ac7232fd7204787ea66a9e928b91cfa57570b5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 19:26:28.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":4},{"id":"58ab31c6a4e0f7869f29bea5491fcb9befc398c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 21:02:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f87099b4c16248869cc5fc21ec2e058c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c2403bbbc3489bbbb41976f354a7ce : SUCCESS in 7m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9fc306070504a5fb68385082f4cda48 : FAILURE in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22b8957aef7441c2b0b35a4ce8fb286a : FAILURE in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a9c04bc91c24717b5a5da30b7e7757c : FAILURE in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a13258a7be4f828ef429a898a15978 : SUCCESS in 14m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/50c2a76c9bab4abc9550c39bede8e220 : SUCCESS in 1h 15m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8ca4e9890fa4bd0b28e8bb1224dfdb8 : SUCCESS in 53m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac40c3547c3b4fe0832aa0d67f0e69ec : SUCCESS in 1h 41m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c42038be5d43c48b1127f65d32c13a : SUCCESS in 1h 01m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5c5396fed16429d812ffb350e34e3ea : SUCCESS in 9m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4bd0fae8a73c48669afe5be1862cd18c : FAILURE in 3m 35s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d2199e145997443a8aa80caf143e98c7 : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/021e78b14e0f467da34f0b62e5b64b40 : SUCCESS in 7m 11s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/dab9983a5e0b405cadb881507a3471fa : FAILURE in 5m 05s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0bf784c8ce944a5bb16c1f0b6f165b45 : SUCCESS in 10m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a097adc23d7943bc83bf0b29f9974181 : SUCCESS in 8m 45s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/91ca9289717c4d21a819ad98e25d0daa : SUCCESS in 9m 56s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/05c95334d46b47108ab6480eb21c2ce5 : SUCCESS in 36m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be511bdf4db84f108262ffbe1dc052da : SUCCESS in 6m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8f3ce3ffa7d4190ac3d093332ddc46f : SUCCESS in 1h 41m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a5b13316848f4ac8b93296d3d9ac12dd : SUCCESS in 2h 08m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/21166c1000274affb69581ec583b71f3 : SUCCESS in 1h 29m 05s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/e48a9a2aae4d4feda193e85ce218539d : FAILURE in 2h 27m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4dfa7b28a30d4a9bad7e27f83639ceea : FAILURE in 1h 38m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/2cf35887c80e40d8b86ff8073a0c0d7a : SUCCESS in 37m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c084484837c54bf4860830f6f1355a2a : SUCCESS in 1h 15m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d2adf57aec4e49e3a7a0d3a70c29ef06 : SUCCESS in 1h 18m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1eef01c6116e4e7f8156236186f60536 : SUCCESS in 25m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74ef0fa911fc4391a5c2bff8da73ee5d : SUCCESS in 1h 16m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84ad7082669a40d89be9edd25d532fed : SUCCESS in 59m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"9aa19d1ce2c3fbb34eb34138fd5c1a87bc80e68e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-13 22:31:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4adf4a62ec064cb59cfe3f498126408f\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5f55f4cdd654277b33180feb23a0fac : SUCCESS in 42m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/74a0cf807cd04c31bab5f66468e26fd6 : SUCCESS in 36m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/736e39cb00324210834b3c72bc96b171 : SUCCESS in 43m 48s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef15d3f96e5a4f41a336313e658821f0 : SUCCESS in 38m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2933f9e109c48369416db7a394dfced : FAILURE in 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ccc9390a371dd234f82fe565ed1caf066aec6f72","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-13 22:48:23.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/813 : SUCCESS in 0h 46m 24s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":4},{"id":"e4b6dd62b2bd53c22201c46b828c42ee463e84af","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-13 22:54:59.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/907 : SUCCESS in 0h 46m 47s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":4},{"id":"06d3f8cf0bec411d8ca209d19adb816f2d4dca39","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-13 22:55:14.000000000","message":"Patch Set 4:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1174 : SUCCESS in 0h 46m 31s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":4},{"id":"55affb115c461a10058eeab80f368baa9acdbd24","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-14 02:31:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f08cae31f74e4114ae5fd92566684e10 : SUCCESS in 45m 13s\n- cinder-storpool-tempest-iscsi-multipath build/562ca5951d9446308b32f0ba683083ac : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":4},{"id":"068fa4572302a290b4d150a8b5ab299fb8ff580f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-14 04:32:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"01569337661a4791facc055ad3d3f93c5067c6a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 04:34:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b830f429aa504155b6e0f7b8248eb862\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc65097d9a8d4cb8ac36a29530e44d2a : NODE_FAILURE Node(set) request 4f8eb6a19251489e84c4013983ef94e6 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0941df5703d94e958075d67fe74dd482 : NODE_FAILURE Node(set) request 4a3cc51436204056884a0822a9cad7cf failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94b890dfa08d45f3b06716f7ec58a2a9 : NODE_FAILURE Node(set) request 9598560806ad485392e1217ff43e8744 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7061a582b0db922134e38b7f29209ba4f2774037","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-14 04:45:02.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"51e66e4d19df738f62090d41ada8dfeaee084a5a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-14 04:49:58.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/22e9d780aacd48c58fba8cf93b8def19/","accounts_in_message":[],"_revision_number":5},{"id":"0f5451434d428d49a0ce53699076b35954231f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-14 04:50:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4c794dcbb1a2029a9d217b16d07c535e8e524a30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 04:51:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/186106c082804fb4bad25e94ef3d4961\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/030ccdcfdc98479ebd0d00c255948f5e : NODE_FAILURE Node(set) request 60012ca61a9148f6a81b99d8c679a504 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/50e4a096d666426e939148a9bc2a6001 : NODE_FAILURE Node(set) request d0f09c1d0fc3491abc0084a9ae49ea86 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5977ce98329d4bb2a057da4d2be815dc : NODE_FAILURE Node(set) request d5319dc57ba44aa6a06a5f8fc85e5c3b failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9c4e14210b22ab02100df445f2fb76efb7c11430","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-14 04:54:46.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1a9f69cd00840898578d0c99dcfaa999260e20bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 05:28:28.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":5},{"id":"fdef06b43b9a7eed2a0ad844f14338148eaaaf42","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-14 05:29:22.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5c836811828448f6a7959134d903dd10/","accounts_in_message":[],"_revision_number":6},{"id":"1d1edd1898ebe3cd620745b38343854c44bf8b2a","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-14 05:42:15.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/910 : SUCCESS in 0h 47m 50s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":6},{"id":"cf38e7abcc05858a63cb2539f199e570c4ab2097","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-14 05:43:15.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/816 : SUCCESS in 0h 48m 50s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":6},{"id":"3f5f6400c608ac6cf286cc382f3f50eb52827490","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-14 05:43:20.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1177 : SUCCESS in 0h 49m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":6},{"id":"3ae16733f18668998d0df9ed82120b0922a092e6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 05:47:28.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":6},{"id":"8fc12c06a678bbd1680405eb196eec8aac7f013f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-14 06:15:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5c836811828448f6a7959134d903dd10\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf9a88d161524b3297f3096c7263debc : SUCCESS in 39m 17s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c47b342451924bfa8d10ece3f4209ae4 : SUCCESS in 38m 12s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b79b1bdb23484b71ae7fdff5c2d615fe : SUCCESS in 43m 22s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dae27a6d39984b9094f8747d898d86d2 : SUCCESS in 37m 59s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2320fffe7d774375bb4c6215ded6c155 : FAILURE in 15m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7d43463fd3323604f1a1f8ac74cef02109d0dbc5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-14 06:29:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/3759a5ec2e6d45009c861072c1d3665d : SUCCESS in 52m 18s\n- cinder-storpool-tempest-iscsi-multipath build/3acea0de985040ea9bbab7412b61fc27 : SUCCESS in 48m 06s","accounts_in_message":[],"_revision_number":6},{"id":"53242bb4f1408d8adaefaf8f8a63b5262163a6a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 07:04:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4944/.","accounts_in_message":[],"_revision_number":4},{"id":"52b9b8904dc0826d204f71b140ca8054cdca35fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 07:32:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86dbb0fda2294c47b883f6c490047474\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be6706b4ffa48d2af8cba9fe9e4574d : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17e1a10613ad41ef99bcc9a9351952c5 : FAILURE in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d34327058ed4ba499db59f52bed6ce1 : FAILURE in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96fc240fa7694c4bbe4ae27a8c6eb51c : FAILURE in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a77cd7d0054d1289675b2ec4f4a207 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/8da6e5390a2144a8b1eb4dcf03856a0d : SUCCESS in 46m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a18c3d0ac655490a9a537fa8628d2371 : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7264f7700f540dab78820e0c9c16c21 : SUCCESS in 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d5a62eca57b4c258c11e4d40a2e8064 : SUCCESS in 33m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0190dd1e9044fbd90560a90d13bf10b : SUCCESS in 8m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/594a054e5d4b4bb98e07a3a76a8a6a3a : FAILURE in 5m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ea370c5121204257a9484fe0c34ce3b7 : SUCCESS in 4m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c7bcafaa354f45c09f55ff9e9768b956 : SUCCESS in 6m 20s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/5f0d9c1221834a7994bbc18089292b8e : FAILURE in 5m 56s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8b882da699bf4156a8cf366d89d9247c : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eae64df9eca14b31b5feb6b57e91502a : SUCCESS in 6m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/48d1203df6a4414abe9d579faa24b2a3 : SUCCESS in 9m 36s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b80561561f08493483c8b8f4a6e2d4b3 : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b25f2aee00f417e9ec22e1d2228637b : SUCCESS in 3m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ccb014ea7834081b746f2e5f8bd0089 : SUCCESS in 1h 32m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c840a642f2de4493969959aad7ae89c5 : FAILURE in 1h 32m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a4747aae05fa401699bcaf9969f33675 : SUCCESS in 1h 19m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/f9e6816bf05c46baa140a702dcf7c5d4 : FAILURE in 2h 40m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c37e91d9a44b46f18f482d6e48475a20 : SUCCESS in 1h 35m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e8f0babcdbba4378af79855f21f47535 : SUCCESS in 18m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23557e0252fc49839f720f77a4f3e123 : SUCCESS in 1h 01m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/acfd11722ebf4c23915ec207c21cf44c : SUCCESS in 1h 21m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ca90b97e62b43a29ca64e33b57a6ca7 : SUCCESS in 44m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f432ef9cdb0f40169a39f66a4d02c049 : SUCCESS in 1h 36m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8888a2cb39c348f294da01cf24916f02 : SUCCESS in 52m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"f295cd8145a38639efeff3e377cadb0ba59e9cbc","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 07:52:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5120/.","accounts_in_message":[],"_revision_number":4},{"id":"bc38e6f084e71e42f762ceec4f623892b3eb0513","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 09:07:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4946/.","accounts_in_message":[],"_revision_number":5},{"id":"6d7d3177cdd5972a41e2b4ddd29290dbbc99394c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 09:54:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5122/.","accounts_in_message":[],"_revision_number":5},{"id":"eacb88ec8dde3150e820941a7ca1f128d891f616","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 10:05:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4947/.","accounts_in_message":[],"_revision_number":6},{"id":"d96eb81a64ef225eb576f2f9e3715f88e6b4d1a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-14 11:02:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5123/.","accounts_in_message":[],"_revision_number":6},{"id":"4a22296f9cd8c2be396a0a5d6a1a847a2e1be46f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-16 16:26:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d878d6a15110cd5a26e7f85245133ce87d1f7ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 16:52:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb8e2a1fc6b46f5b334ba98a7e36a15\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a14a955612d47138d7913499ab9b907 : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/be7c66cbba3b4f79bd0cc095f9b51257 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1bbc930839f4ed0841540aff00a7bba : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a0b98fc71dc1303b1f64a7adfc88cb53f95c457b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-16 17:27:33.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":7},{"id":"b4b93cf7aa3d486e1e779c354c89aa728855e712","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-16 18:44:50.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/8222e0965ca342aaaa15316bba4fa2cd/","accounts_in_message":[],"_revision_number":7},{"id":"6bdc7d55adc0d4a337ab8f65a6fd9398e7322eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-16 19:06:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2acf912bf214acfbc766b5361fce072\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf4daf4050249468a87442bbb517b82 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07cd9d5b7596406fad5fa8451652dc1f : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf858b1e23fe425886cabd6ed06541a6 : SUCCESS in 9m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64ad6897bf51485780555b5e199ce06e : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0837c5af468243fe8d4d764a963168d5 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/0a4bb71fe7ba488a8d37d38f0de8f92d : SUCCESS in 41m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38fcb0e20cdf4539bdfc94b0877fed95 : SUCCESS in 59m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1bba3a47d10469dbc043812cf63c190 : SUCCESS in 1h 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bd18738afc64e00baab199c78c51bca : SUCCESS in 38m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53cb26cf521d4344948838990ecaee72 : SUCCESS in 8m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/71d4942f9b9f4ab9b3441d6a1bf83032 : SUCCESS in 18m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c64885ccb7f44ba190c7a6059b663c1b : SUCCESS in 3m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b43655acec94c2abcb772c12d098a04 : SUCCESS in 7m 08s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/c1b364cd55f64751a0e5c50d294b98c7 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32489cb7011b40489ce1fc541b909b09 : SUCCESS in 6m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a8853af974f64816a92869056f2a3e4c : SUCCESS in 5m 55s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/05b5d8957d0c4a5686d6901e03829ded : SUCCESS in 8m 05s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d1b8d9ecc69b485da5969bb63d055cbc : SUCCESS in 34m 42s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ff30a55dc3345538d7acaff07bd8c9f : SUCCESS in 7m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3832efce9ae840038cfb1fd59dd40578 : SUCCESS in 1h 18m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d94eb83d3361417b8618d28f61c9f824 : SUCCESS in 2h 01m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57a53d55e31540fc9a60f45f95e1f897 : SUCCESS in 1h 19m 08s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3a96d6b11cc746eb803b2097e64dab6b : FAILURE in 2h 37m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/abc00c218ec84649a19d7d8a2cb9c6b7 : SUCCESS in 1h 50m 07s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/703857422d794ab298e7cdbb7090e612 : SUCCESS in 19m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8e27ff2cefeb423792a640b90c72cd30 : SUCCESS in 1h 15m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4f7d7b1cea9245d892c1cf3619b9ef8c : SUCCESS in 1h 26m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b84734b5e9e4fb58aaddc38c56dea49 : SUCCESS in 43m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1418423698a54222acacd2f45d25434d : SUCCESS in 55m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de4a8d79904240ddb97037c42e9d6c26 : SUCCESS in 48m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"0e886640e793fcf908183bbf63f6d9e707e8c16b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-16 19:21:28.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_08_2026_19_14_26\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"cdc725bc20569a7d097ea45b9c9f984bbf0cb6a9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-16 19:33:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8222e0965ca342aaaa15316bba4fa2cd\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d044137ff7bd442ebe9ecf15203f566a : SUCCESS in 42m 42s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/69b686f1a5b94ea991c1515a9884d44e : SUCCESS in 34m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4322ecd9b9a44adc856fd0af30edd602 : SUCCESS in 53m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de2706c3b96f4b38bb1083a6dee2a7d5 : SUCCESS in 34m 21s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9df5e86fe4504d6087b82283754c5165 : FAILURE in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"35f89c931407693f1765ad52d9f514dc3d8e494c","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-16 20:22:26.000000000","message":"Patch Set 7:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/834 : FAILURE in 0h 8m 0s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":7},{"id":"0968a8679e267ef24d80f3025dc72ec42587d29e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-16 22:21:06.000000000","message":"Patch Set 7:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1195 : FAILURE in 0h 4m 5s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":7},{"id":"b23262735ef978b1fec9dd2967d526abf06ae843","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-16 22:42:08.000000000","message":"Patch Set 7:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/928 : SUCCESS in 1h 41m 5s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":7},{"id":"c36833a86f6513116539539c5adf498dfef13117","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-17 00:04:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/8250269db4cd4130bf98675a497c5ad9 : SUCCESS in 45m 46s\n- cinder-storpool-tempest-iscsi-multipath build/4f5b68fab87f44d1a47202b7ca030aa2 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":7},{"id":"716cb5d283628884e413dc4c9706b00f08923fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-17 05:48:32.000000000","message":"Uploaded patch set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"b0fc5267d66ec301369f8245410e06a5eb0c694a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 06:06:37.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/2eead9ac24ae4e9e80b6544542613649/","accounts_in_message":[],"_revision_number":8},{"id":"6ad51daf63178cea603d6650a9c1bb530e7dfb3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 06:15:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5636ed3da88c42cdbefba1612ba094f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc58976d808c4fc5946db24326418d30 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/faf67ba92f394357b270bacb00023199 : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f93b520a24e4013bf78ee84867e79ef : SUCCESS in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"12a1c8f8235c7c73d6f15bc8e5cc72e40b7601f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-17 06:40:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2eead9ac24ae4e9e80b6544542613649\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe8018638bd34e9cab5ee82587cb0830 : SUCCESS in 39m 49s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71384ee771c84f15938149fe883aefd3 : SUCCESS in 35m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/15d87964d0614200b48ff0c11f65aa17 : SUCCESS in 51m 32s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff970d58f62643adadc17607a63df35a : SUCCESS in 32m 22s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fe18ab2ecc8c4f5298724d921158dec5 : FAILURE in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"aeab0c735d3329172788beb14fca2a515baa6b42","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 06:45:41.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":8},{"id":"f2bcd8657442bbc49033ce381d89ecbb7b416b59","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-17 06:46:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e3d74d89be294014ac30b42684286cb0 : SUCCESS in 42m 40s\n- cinder-storpool-tempest-iscsi-multipath build/cd14410f1d1347bda937e7455fe2274d : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":8},{"id":"2a287ae75ff6cd182061780364fa25c7ade5590a","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-17 07:51:13.000000000","message":"Patch Set 8:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1199 : SUCCESS in 1h 21m 0s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":8},{"id":"641e0e2d372e0d0f421ce9257461c76125c7809c","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-17 07:56:50.000000000","message":"Patch Set 8:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/932 : SUCCESS in 1h 18m 33s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":8},{"id":"47f6fef01f972dba8e178a3d36c4b166dcccd944","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-08-17 07:57:18.000000000","message":"Patch Set 8:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/838 : SUCCESS in 1h 24m 12s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":8},{"id":"3a174e6320e894123461cc980567727003540d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 08:29:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dc36729edd8411990645a40ce01c3ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15f767ac1314202b18365f31d23ef01 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/487ef5147d3d456ab91e8ea3ebbe1aff : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7e7a937c7ab43dcbcef6faa7d23a246 : SUCCESS in 7m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ce6d8fbf15d4e91a4a47cc9bf7eb815 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ff284ae27e442db46f90596c32c3ea : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/29546a406dfd4c3e973f992985df2941 : SUCCESS in 38m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02157af5398f4af791f1966632ee1d15 : SUCCESS in 29m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7dbc7acc0d8474883db424c03157a44 : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a67cce3e171645eea518998c3b31c37f : SUCCESS in 56m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ff012b28d124f6ca644aaa3c88b349e : SUCCESS in 7m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9897268d8b9140b6b2568dbde4cd15f4 : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/37be98acdd7b4b748d74daeca1bc6cab : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b58fbef57b741e8be7ed7899cd59589 : SUCCESS in 5m 16s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/51acbdadd9aa4d4b8d4ae10ba8fa0edf : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/aa544147de014a7592226c046e195759 : SUCCESS in 4m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/51acf6a5a14c4cb3bfee041d268e27f5 : SUCCESS in 5m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/15be84811ffd430f9b9a2a974a5d8195 : SUCCESS in 10m 20s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/87617877580f42adabf812f36198e0b7 : SUCCESS in 38m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4eca5fca90f14d119518f248eab5e013 : SUCCESS in 3m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/294b978e4f9c4c89ba0dc5b3e456635f : SUCCESS in 1h 32m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/770caa10cb6b43c0aa988f7a2d5db622 : FAILURE in 2h 04m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ef5e4d6d18f44eea5039472c676ce01 : SUCCESS in 1h 31m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/2bb8a78abe474b34a7e1604849bb1ab4 : FAILURE in 2h 35m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/94c50a146ed74eb39ea9e9005d887d14 : SUCCESS in 1h 24m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/43b7fd72ac604ad9aaea92f798e2ed3f : SUCCESS in 24m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c738deef95e433593199079dfa7a83b : SUCCESS in 43m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c94ca1ab86d24c7596e1e353e8b73acf : SUCCESS in 1h 36m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/904a6ce8736445298c75f7d0eee44e7c : SUCCESS in 31m 37s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01c664194e12439984de370f7ef85edf : SUCCESS in 1h 29m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a66b14685d47ec8024e4b19101445f : SUCCESS in 31m 35s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"750a5ff6ff9a44f2535f80943a4b9629380f2936","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-17 08:39:55.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2026_08_33_10\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"d9136374b2bb63e8f0530c8e798e3f1e37778456","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-17 09:33:17.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-7-fc : FAILURE in 11m 31s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-7-iscsi : SUCCESS in 1h 12s","accounts_in_message":[],"_revision_number":7},{"id":"9601a6e6695156f690a7e5f6f634b1335d99ee20","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-08-17 11:10:26.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c21a5b0681d47d163f9e0090bcf50889d36c60a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 11:38:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4965/.","accounts_in_message":[],"_revision_number":7},{"id":"cb90dbfe3b09f7ac059a2d5e62970c81afe1d3a1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 12:40:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5141/.","accounts_in_message":[],"_revision_number":7},{"id":"a110f02a3cc242f047986cb6abe51ed71664bd0c","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-17 15:38:24.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-8-fc : FAILURE in 13m 29s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-8-iscsi : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":8},{"id":"b0bf5c1ec904a04d4bc1b1f09479ced8a179de53","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 17:13:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4969/.","accounts_in_message":[],"_revision_number":8},{"id":"ed6d95abd66d65445bc3bab6eb9ee52e97f85086","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-17 18:39:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5145/.","accounts_in_message":[],"_revision_number":8},{"id":"a919217a4dc57c8cd19046ce53e9619319a6eee6","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-26 22:36:19.000000000","message":"Patch Set 8: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3565d1ce6c9032a9c2429ded09688f85d4dfbdbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-01 08:48:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"95c5764abbfc4991e5b25554bbf03ddf3f3f2a7e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-01 08:58:10.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5347/.","accounts_in_message":[],"_revision_number":9},{"id":"1827ff9822e17deed65db6ca25502e720167739e","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-01 09:12:16.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"920284ac8bffa1610701c8b19c97da5516c77c74","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 09:12:58.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/38b9218d865e4ad9bc9da0ce49d2c6e1/","accounts_in_message":[],"_revision_number":9},{"id":"52369699b4f0994da0d350cd364feae3ca0fbe53","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-09-01 09:36:26.000000000","message":"Patch Set 9:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1123 : SUCCESS in 0h 47m 0s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997422","accounts_in_message":[],"_revision_number":9},{"id":"88e59aae8f2b47ba9272e0d60a6f5c704996014b","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-09-01 09:36:37.000000000","message":"Patch Set 9:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/1031 : SUCCESS in 0h 47m 9s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997422","accounts_in_message":[],"_revision_number":9},{"id":"673c645eff67ba0014e8039defdb3aca180273e2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-01 09:37:27.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/38b9218d865e4ad9bc9da0ce49d2c6e1\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dacd5dc28d104aa995672505efcb8033 : SUCCESS in 42m 02s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fa592b5252b340cfbaca97e8ecbaed08 : SUCCESS in 36m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81a9ea1be9a741e5b67c4d1a9729bf21 : SUCCESS in 48m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/15a82de193214c94bf108217c4ee28ec : SUCCESS in 33m 53s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a84d3666c21b4531860b5ac0a8373aae : FAILURE in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"549de40e1ba61522ebdd9a866855ea2b4098b3ca","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-09-01 09:37:40.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1392 : SUCCESS in 0h 48m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997422","accounts_in_message":[],"_revision_number":9},{"id":"031a4adce5ede835a3e938fcfb490490f0a6a1e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:41:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac346dda3fb4a14885da3e026f89ede\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0c5f702070145ea917eb2b0cd9edd8c : SUCCESS in 16m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35e133e0069a48bea0c9179133024c1e : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/384bda0d58ad4fb2bf005a6165a009d9 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aaa866c218b8a5ca9246dd33b50ec8c1c43abc2a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-01 09:41:24.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ffd1196177e4445da2286c8f59c30eb0 : SUCCESS in 47m 40s\n- cinder-storpool-tempest-iscsi-multipath build/18a700db9f874abd946c35863cfbab00 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":9},{"id":"6309de4517b609742df94cb6de3bca697de80fec","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 09:43:45.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":9},{"id":"225556ee4407212c78b39a5a8144b00d64e1cc72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:54:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cc8b3c8bee54c538eb12d2f57e719db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad63e05fb19459ab9dda092dfabda24 : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a8a8937d5d5485eb6463a106aeefac6 : SUCCESS in 12m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfb134bdd31e4aaea637fe374e36e3c3 : SUCCESS in 8m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5ba8ad1ad3241d28ccc8086d7585b7e : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3c026cd8cc45efb1b3cb5bc6a876f0 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/7fb6fbc224d04ec48dabcf84e6590a95 : SUCCESS in 47m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4474674c267f48019f9c9c591d98c2b1 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b1cd71fa6ea4a0f8cb54b62279aec21 : SUCCESS in 1h 23m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b0e112f6823494a9d4607a8772f8c5b : SUCCESS in 46m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/814775d200624153a542de3aa4524ad1 : SUCCESS in 9m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d9040ed2bb248ef859842a98cf771c0 : SUCCESS in 16m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fbdbe6857ec4494d8b220577d2037825 : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b0ef16c60404801a413de3a45bb7c27 : SUCCESS in 4m 05s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/699d996a24df4ae1a061d323a5b4514f : SUCCESS in 10m 52s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e49dc6e6afee4ca081909001807192dc : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b7aa38c028bd4c7aa8350127f4381aa0 : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6002c83e1dee4f78b9d3885785f06105 : SUCCESS in 5m 42s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/efdb7954dba548f1b8d4cb5a4ab120b3 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/dd3fb7c4823f4c8b97e291bb1368fd49 : SUCCESS in 7m 07s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bb633d1dbd78436086a86cc5ed6e99b6 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc172ea33fc547479b64f853bdbbbff5 : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7326d37e0914add9335214691d9657f : SUCCESS in 1h 28m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c17053ae50b64a4183c1320892bf7a6f : FAILURE in 1h 03m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7cbab8ae8a814cbeaa6aab35838e1ddf : SUCCESS in 1h 58m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ebbf1b42d1d746b184af1ab66a31b7a0 : FAILURE in 1h 35m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/4359a138372e41dfba8840d423abf46d : SUCCESS in 1h 24m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/fc6423d81b3a484397303aa55dc3d1ed : SUCCESS in 35m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47d2d175d8874c22bd3d02df0e218378 : SUCCESS in 1h 20m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d071603759c846ca84240376b26183b9 : SUCCESS in 1h 32m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/72f1d3ee020a450d862ad6573d2cc3fd : SUCCESS in 46m 13s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a7f4afaede594e68a0c2ca8eafed3256 : SUCCESS in 41m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e36facc76b84949bbe65ce34ab36938 : SUCCESS in 1h 02m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"4b5d7a1d7c46f76236b18be05881ed013745166b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-01 11:04:58.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_09_2026_10_57_49\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"93b1f5c7ffc722c3cddd42014f7328b9dc74f976","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-01 12:50:58.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f12de28ea5a49bf035a1f328b904a4cf58dd77ef","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-01 18:52:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-9-fc : FAILURE in 12m 40s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-9-iscsi : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":9},{"id":"29741d0b543edce800e5c9ea08cb97159a204a8a","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-02 02:10:51.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-9-fc : SUCCESS in 53m 18s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-22-997422-9-iscsi : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":9},{"id":"dbbeaff2d8ff6117a5307c56a8a0570096f04444","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-02 05:39:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"165dc6c8ce7539f8236a8994f53f378fc7e2231b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-02 06:09:46.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/997422/9/upstream-check/cinder-tempest-plugin-ontap-fcp/a276264/ : FAILURE in 25m 10s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/997422/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/04adee0/ : SUCCESS in 26m 44s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/22/997422/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6701f47/ : FAILURE in 28m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"8c705322c31adceab988534a0aec7fd319eab5fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-03 14:06:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5171/.","accounts_in_message":[],"_revision_number":9},{"id":"fa3a8d6e7a9f897c12ce34f899d2642d12949cdc","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-04 06:36:51.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"be79cf7a0363e29a36cb6469168ee2e6268b7718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-22 03:32:38.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"0fd6eae3c79e137ecb9e0ae514c267732aa78cfd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-22 03:53:28.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/3aaa11262f0c47e4be4ec19b37a73d04/","accounts_in_message":[],"_revision_number":10},{"id":"02cef66f3866156b7d8db55a4530488e4698b4fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-22 04:26:02.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2 .","accounts_in_message":[],"_revision_number":10},{"id":"cef3095c3d64567aca3b17208b4194228cc4d599","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-22 04:40:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/c1b93bd3482a4b7897f96961b91e57bc : SUCCESS in 45m 59s\n- cinder-storpool-tempest-iscsi-multipath build/777a354bc8814aa1b0ad464aa195b739 : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":10},{"id":"0cf14ac07defb9e9345f1a01d6d7c8159deb6ee1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-22 04:44:27.000000000","message":"Patch Set 10:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 21s\n- FC       : SUCCESS in 0h 47m 38s\n- NVMe/TCP : SUCCESS in 0h 46m 25s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/997422","accounts_in_message":[],"_revision_number":10},{"id":"c8e874870fc7d3912f59c70966fba96673b50984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 04:58:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27825a10de424253b8f2ebb55574c19a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bc25d863fcb4bfe96d78f6a10cf0189 : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5cad2716fc6240f19053d8eb92022e31 : SUCCESS in 22m 55s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3dde0d99d7c64ce98c04b06e66d077f0 : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8b7810324c9ab9fdfc7a41e851b5d9ace32e9c60","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 05:26:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5623/.","accounts_in_message":[],"_revision_number":10},{"id":"3f04e527ff9ef9df395f7a917769573c50278e0d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-22 05:28:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5446/.","accounts_in_message":[],"_revision_number":10},{"id":"2c1c6cb2d9c396d64a565b6f04be603c53bdbaa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 05:56:04.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed4648616d67481e887048829287f1ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb243e0e600e4b07850253f1246886fd : SUCCESS in 9m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b98e532a551b48d3a864683756599d84 : SUCCESS in 12m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a546af94040c481784841802668c8fd4 : TIMED_OUT in 40m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df5930058f084b8f8d33259c1f0932ea : TIMED_OUT in 41m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78433c9db06f42c48d7abb203bfbed25 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ff229a7d4f8242b18c1c2b98c86b0cf8 : SUCCESS in 1h 06m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/293cf8547c7546a7b4b4119eba239f1d : SUCCESS in 1h 03m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e671febdb5404969a7e421ea8da303f1 : SUCCESS in 1h 26m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecbf06ef92c74b8c9a8d54b9d1b16e89 : SUCCESS in 59m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9d953128851405881667a15eea3988a : SUCCESS in 5m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/47f290aae04d405397f697f339e4f8db : TIMED_OUT in 41m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/01a1af7f7382481c837fb0a484d68e70 : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6bed20bd2c0e4a4299ab6273f53a04d8 : SUCCESS in 6m 04s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ac313ccad4564773b374d4762d8a0257 : SUCCESS in 14m 16s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/083c983815b24609845c86bce5405b48 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/31e6621b42b5432da21489e32deff8f3 : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7bbe26c057ef4aee87120d661b1a788e : SUCCESS in 6m 36s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f3c5cc1c404144f49f8e8c7cb1f93cc9 : FAILURE in 6m 16s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0b62092838074be881553a6a06ef6776 : SUCCESS in 7m 06s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c0f394fe96df441cbc8c3fe801b54abe : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7a80dd22f26a4958a40f4de938a94cb7 : FAILURE in 3m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f2c9de0796bb4d51bd727654a88b08f3 : SUCCESS in 59m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ba719722e294a7ba9c824da0d4e0df2 : FAILURE in 1h 50m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45c3286c9e41476a81f584b22c160874 : SUCCESS in 1h 53m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/99061bfd0dd5424ba62ae434d95b8f05 : POST_FAILURE in 2h 20m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/73c6ec2ec19a44ad9a2b9f7a82a96add : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/859e3b2a9dd04de1bf7c3adc25ba89d1 : SUCCESS in 27m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d49140ab22e747adb9b2fcc50f7b3ac9 : SUCCESS in 1h 19m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/04b0a4a6497149779f4b296365b3b03b : SUCCESS in 1h 35m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2195e3f5771d4fbea09d9d1f1fb819cf : SUCCESS in 49m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a6f8c9ce16843babc9a70dc3ce034c3 : SUCCESS in 1h 35m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc27b5dae93049cc812d3eee897cb1b1 : SUCCESS in 27m 38s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"81d931a03c1d066ec2fba9d0ecdb89d053629339","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-22 07:55:43.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3aaa11262f0c47e4be4ec19b37a73d04\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ab28e736ed9640eea056ffb70d0c678b : SUCCESS in 47m 43s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3596c3f6f466463a9ad1520e51dde9c3 : FAILURE in 2h 43m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2391fc4e0f614f06b562200fabd25a4c : SUCCESS in 50m 06s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e5b2a088f69148eaaf0ef32727ed49e0 : SUCCESS in 38m 30s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2b358bd9d3b842bfaf587d765c377783 : FAILURE in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"05bc0cac8a1a84483c5aa9ff65dfd59bc84dcbf5","author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"date":"2026-09-24 05:22:16.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"74d1c583529762a9e6a355adcbf15e1f73f6f7c7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-10-02 14:10:41.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ce2d9562f93a62cc856aa41b0860d21d3fee54c7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-10-02 14:26:56.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f9f40017a8245f9555ab35bb946ff86ccb8a7578","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-10-02 14:27:11.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c499c2625e848b6211f5334e1837e380bc1174dd","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-10-02 14:31:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"33bc25c92b7686eb39e6f42d09dd6b51620f45c9","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-10-02 14:47:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"467a719237b0c32a0908ef97270778e3642c5949","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-10-02 14:55:15.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"67e0a873a75a484c5f1d734a4899e40f9d6a2acf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 15:03:59.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d66a38980c8c4e209b4b49439ab4ecf2/","accounts_in_message":[],"_revision_number":10},{"id":"c1ffaa4a54dac49496556d5cb21bd69d76953b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 15:08:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde615b465d44e6facbdfbf11b60ec61\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eccf5bdb85054bb9a4de77510b2b1ab6 : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3110d82faf8b44888837bd96d082f6df : SUCCESS in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d8c5d1c95b7e9c71a53e7c8ae6cd1b58228ca4fb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-10-02 15:17:55.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7ceaad8d2f4819524471f21acc2d3e8320ac0b09","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-02 15:47:06.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d66a38980c8c4e209b4b49439ab4ecf2\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/498bad0cfba54c82af80b0ef1c91984a : SUCCESS in 33m 40s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6fedb46d0fb54eda8b84b6a37c486f24 : SUCCESS in 34m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/71faeb02d5ac4ef58ccd747632f756da : SUCCESS in 39m 06s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3a1259f8371426fae2391d205eb9de1 : SUCCESS in 28m 30s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5814a416feac47a4859dda3035c721a2 : FAILURE in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ba7e61b67ddf76b6923d2df2add6b2949d5d86e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 18:03:54.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d791ec62472c469398be22ccfa11211b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549e0eec902f47c5a663bf247d93ebc6 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b660f0f3b91044faa0e2c9dee88402be : SUCCESS in 10m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7955d00bc1b842668f6ccdf03d93f97e : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f2e782fb724c8fbb5476dfad014916 : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/babcfab9cf5f456dbcaa09426d0bc012 : SUCCESS in 40m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfbe0fc4748c40c19b0b05a044a31ef9 : SUCCESS in 1h 03m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/456e98702d144ad484f01271f1f69ab4 : SUCCESS in 1h 31m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45a755954e214462870bf670c4eab279 : SUCCESS in 1h 04m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0bdf279b39e4be785204033e9acf09b : SUCCESS in 9m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/093521f4a58246d89ea1e98bfaae9116 : TIMED_OUT in 41m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5a9b56b7d7f4dd4b713c54d58505bf7 : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e86094ba04e643dbbf5c4b366ade15f6 : SUCCESS in 6m 37s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/4b27d4f0c73f47c88b2cfc4466961bbe : FAILURE in 12m 02s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/50f538afb82b4ba89dc38e48444ee085 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2c4a983cad1d4a589e709943ff887b21 : SUCCESS in 6m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/719fd3ae78bb4f108b91480ed9a84fae : SUCCESS in 6m 03s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/212dfcaafc274fa2ba70936052fc5150 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/3e1ab3abe39a41038ba9df402ba62c70 : SUCCESS in 10m 17s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a027ee05154646fd8a121dd8a5eb4b96 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9212b3f92e164bb0b0014fd0ba434aa3 : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/acae13df22b24f59962206e2113e9710 : SUCCESS in 1h 34m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c1adb74d32a34a098728df8102126251 : FAILURE in 1h 55m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dd78dc6e9b27434a99336bc134279693 : SUCCESS in 1h 49m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/62a714e2642b457c8d78df6ba277f495 : FAILURE in 2h 51m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/b96244f5e1504896abfeeea9bdf9850c : SUCCESS in 1h 33m 56s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d117cf061c6543b98a4eb025170e0b61 : SUCCESS in 33m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a86ca97f33124583b9cf721aaa0ebf12 : SUCCESS in 32m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7eb09f66d7494c5da1e01ef4273fec0a : SUCCESS in 1h 38m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/34a651749b7b4e9a8793a1277de7fd6a : SUCCESS in 27m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a767a6085cb244f4b2626470099edde7 : FAILURE in 1h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1149f342512a44608f4265e8a45969fd : SUCCESS in 32m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"4e987730f944185f9f5856c5a8eba6f57a9bfacb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-02 19:42:55.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/b6bce60798164b5d9c9b1e63da75bfba : SUCCESS in 43m 43s\n- cinder-storpool-tempest-iscsi-multipath build/c44d178a2e63421bb2dfd78d0d27d98c : FAILURE in 1h 15m 48s","accounts_in_message":[],"_revision_number":10},{"id":"1f8c65ecbb269c726222e0e9151776aedd10b0b0","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-10-05 04:13:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"abab3c035a0a4ad501110d98268302eca82b333c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 04:29:04.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5c9d4f252e024078b3673d574ec01b65/","accounts_in_message":[],"_revision_number":10},{"id":"0bdce1ae4a81cd1224022fe4d3dc37b39fdcc797","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 04:38:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/434776ce7cb447cab961736bbf31941b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9566a5d61ee4e00968eb618193560e5 : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/894cc7cd385746f1b2bdbbc8af2b6066 : SUCCESS in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"73b8ae8c93719224a15e33beb8a265c400645d96","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-10-05 05:00:00.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5c9d4f252e024078b3673d574ec01b65\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7a9ffc2d1db743faac0f72f297d0a355 : SUCCESS in 46m 06s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7015b985cb3f4d1595fc088d11e9d9ee : SUCCESS in 31m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b7bc3ab229e74a12ad9d2e824a8d9704 : SUCCESS in 39m 06s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76429c05ac8a4ed980a69349328a4342 : SUCCESS in 32m 41s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f204ebfa3743476487f775eb4da554d1 : FAILURE in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"60502ee71bfaf8afe31a99b2afb6e2dd77684b78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-10-05 05:27:13.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/60b44ee7e2bc47929bc0af92e3ec8bec : SUCCESS in 44m 07s\n- cinder-storpool-tempest-iscsi-multipath build/045bb53cc0c44ad7bca27937e9696690 : FAILURE in 1h 12m 33s","accounts_in_message":[],"_revision_number":10},{"id":"357c1a43c75aa58d246b0db2880ca9ea15b5ac2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 06:38:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578fd3b1e9eb4231a7d607471729733c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5098eceddc1743b498ec7cb4babad112 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5978947e35bd4dc6bdb0d7c6e25b3cfa : SUCCESS in 7m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b1273c1864e4c0db3e8465b33f6af16 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a485d8ae8685481c82f5e5b027da2e85 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/84d82cfd166347ff8e8128f8a34fb548 : SUCCESS in 29m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31661585c8744a268d0bed9afb402e0a : SUCCESS in 29m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77a351487b244197a1d191deb935d747 : SUCCESS in 56m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/756975ab71ff4530a3d2ea59b6fce795 : SUCCESS in 28m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b3230f6ef2e4069925eab41012dd293 : SUCCESS in 9m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a0178e5d5abb457e8d29c8259ca29ebd : TIMED_OUT in 40m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fec5966a79474c8f9c39f54b13c5e9ce : SUCCESS in 3m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5a00202205b64f5496857847faca9087 : SUCCESS in 5m 51s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c11d1ba0e32244b5bd6ccafe7824f8d5 : SUCCESS in 10m 44s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/609e50b6fa2048a895eb7fefc05b030c : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4e8099e3c32f4f299fa9bd4425dbe54c : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b43d82a9b20c48c9a02acfbfc1753646 : SUCCESS in 5m 31s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b17a5af34e454f1da5849a42e497fdc1 : FAILURE in 7m 11s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/70b40b7ea8564d1795454decadbea7a9 : SUCCESS in 8m 27s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/afc30eeed3a64b13a04934660ffe58d7 : SUCCESS in 37m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aadf51ff150b47a9b6d2b3bdaaec8877 : FAILURE in 3m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/16c5ae3eca8f4d9b950cf1ea2aa4d985 : SUCCESS in 1h 36m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/eedaacc3ceb64fe3b2856cd7c07919ee : FAILURE in 1h 59m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fa0fa1dfee97475db6a39aacf30ba36d : SUCCESS in 2h 05m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/3a90c952c65e4c0ab0d6e1b138257c85 : FAILURE in 2h 21m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8ce2615484c5454b93d2af2dd175a2a2 : SUCCESS in 1h 17m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5584ee1bd6a3412caafaf0613bba7f1c : SUCCESS in 36m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5d056cb33f9c45258ebb454a6c60d49e : SUCCESS in 1h 10m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4e9944b9492b4a518f275e2007561560 : SUCCESS in 1h 09m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b480f0b937a4baa97e7a9681520515c : SUCCESS in 47m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ccc6cbb616f54bdf872bb2a5b0dccde3 : SUCCESS in 1h 35m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf83cb47c8854998aa094d57a298913b : SUCCESS in 1h 03m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"bc7af6e64f45f88ebce67d8bf4e91ab7dd892d9b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-10-05 06:48:07.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/05_10_2026_06_42_07\nLeave a comment with","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"dec6701c8404ad8e0095ee16e3c960dfa7403dc9","revisions":{"3f5e18cce17395fbbb9d8853a6950a61384e72ca":{"kind":"REWORK","_number":1,"created":"2026-07-15 16:15:30.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/1"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:15:13.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f5e18cce17395fbbb9d8853a6950a61384e72ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f5e18cce17395fbbb9d8853a6950a61384e72ca"}]},"branch":"refs/heads/master"},"356cfa2b68c1cccc38a0f55e0f13f7c16faec53f":{"kind":"REWORK","_number":2,"created":"2026-07-15 17:10:11.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/2"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 17:09:51.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/356cfa2b68c1cccc38a0f55e0f13f7c16faec53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/356cfa2b68c1cccc38a0f55e0f13f7c16faec53f"}]},"branch":"refs/heads/master"},"0bc5ce4ab8ba2016cbfb8aed4e68300e8f7f893e":{"kind":"REWORK","_number":3,"created":"2026-07-16 13:56:04.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/3"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-16 13:55:48.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bc5ce4ab8ba2016cbfb8aed4e68300e8f7f893e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0bc5ce4ab8ba2016cbfb8aed4e68300e8f7f893e"}]},"branch":"refs/heads/master"},"c5c2c992a7f44df9ac128f7387a8e543dd89831e":{"kind":"REWORK","_number":4,"created":"2026-08-13 18:32:02.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/4"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-08-13 18:31:42.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5c2c992a7f44df9ac128f7387a8e543dd89831e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c5c2c992a7f44df9ac128f7387a8e543dd89831e"}]},"branch":"refs/heads/master"},"df921f4a7318315ca5195c8705a138d05ae28da2":{"kind":"REWORK","_number":5,"created":"2026-08-14 04:32:57.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/5"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-08-14 04:32:40.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df921f4a7318315ca5195c8705a138d05ae28da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/df921f4a7318315ca5195c8705a138d05ae28da2"}]},"branch":"refs/heads/master"},"337861b2596b430447f06b91d5370a527fefcf22":{"kind":"REWORK","_number":6,"created":"2026-08-14 04:50:52.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/6"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-08-14 04:50:37.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/337861b2596b430447f06b91d5370a527fefcf22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/337861b2596b430447f06b91d5370a527fefcf22"}]},"branch":"refs/heads/master"},"afb3a122a92e0c4e55c9a919869bfda5c29dadc0":{"kind":"REWORK","_number":7,"created":"2026-08-16 16:26:53.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/7"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-08-16 16:26:36.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.13+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.13) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afb3a122a92e0c4e55c9a919869bfda5c29dadc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afb3a122a92e0c4e55c9a919869bfda5c29dadc0"}]},"branch":"refs/heads/master"},"1ac800957efa7eead2c9a6d91fd846e5efe33b95":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-08-17 05:48:32.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/8"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-08-17 05:47:06.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.8+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. A stale value is\nreused for up to 10 minutes before falling back to the default\nutilization. Unsupported backends (ZAPI client, ONTAP \u003c 9.8) keep\nusing the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there and the performance cache is\nalways refreshed, regardless of this option.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ac800957efa7eead2c9a6d91fd846e5efe33b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1ac800957efa7eead2c9a6d91fd846e5efe33b95"}]},"branch":"refs/heads/master"},"33ccc1575e443c6d0d9a360bea05b8c0a9141813":{"kind":"REWORK","_number":9,"created":"2026-09-01 08:48:55.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/9"}}},"commit":{"parents":[{"commit":"832a8c4c99c9bc09034c0bf4a98670f70d2297db","subject":"Merge \"Set worker\u003d1 as default in service.serve()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/832a8c4c99c9bc09034c0bf4a98670f70d2297db"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-09-01 08:48:15.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.8+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. If a metrics\nsample is unavailable, the default utilization (50) is used.\nUnsupported backends (ZAPI client, ONTAP \u003c 9.8) keep using the\ncounter-based calculation.\n\nWhen this option is True, the REST client is used regardless of\nnetapp_use_legacy_client, because the node metrics API is REST-only.\nSet the option to False to keep the client selected by\nnetapp_use_legacy_client and the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there. The option still controls the\npublished value: True reports the metrics-based average across\ncluster nodes; False reports the default utilization.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33ccc1575e443c6d0d9a360bea05b8c0a9141813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/33ccc1575e443c6d0d9a360bea05b8c0a9141813"}]},"branch":"refs/heads/master"},"dec6701c8404ad8e0095ee16e3c960dfa7403dc9":{"kind":"REWORK","_number":10,"created":"2026-09-22 03:32:38.000000000","uploader":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"ref":"refs/changes/22/997422/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/22/997422/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/22/997422/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/22/997422/10"}}},"commit":{"parents":[{"commit":"e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4","subject":"Merge \"api-ref: Explain why consistency groups are deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e5c1549f4d550d0193bc5f01f61d4b6aba0a3ad4"}]}],"author":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-07-15 16:04:40.000000000","tz":-240},"committer":{"name":"jayaanan","email":"jayaanand.borra@netapp.com","date":"2026-09-22 03:32:09.000000000","tz":-240},"subject":"NetApp ONTAP: Add REST metrics-based pool utilization","message":"NetApp ONTAP: Add REST metrics-based pool utilization\n\nAdd a netapp_use_metrics_based_utilization config option (default\nTrue). When enabled and supported (REST client, ONTAP 9.8+), the\ndriver reads processor_utilization from the ONTAP REST node metrics\nendpoint instead of the counter-based calculation. If a metrics\nsample is unavailable, the default utilization (50) is used.\nUnsupported backends (ZAPI client, ONTAP \u003c 9.8) keep using the\ncounter-based calculation.\n\nWhen this option is True, the REST client is used regardless of\nnetapp_use_legacy_client, because the node metrics API is REST-only.\nSet the option to False to keep the client selected by\nnetapp_use_legacy_client and the counter-based calculation.\n\nThe ASA r2 disaggregated platform has no counter-based path, so\nNODE_METRICS is always enabled there. The option still controls the\npublished value: True reports the metrics-based average across\ncluster nodes; False reports the default utilization.\n\nAlso fix FlexGroup pools reporting netapp_aggregate_used_percent from\nonly their first aggregate: the capabilities library now keeps the\nfull aggregate list for FlexGroup volumes, and both utilization paths\naverage across all of them.\n\nChange-Id: I3ee5c28a3b1be36cb9531b4ef86efcd5c75e89a2\nSigned-off-by: jayaanan \u003cjayaanand.borra@netapp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dec6701c8404ad8e0095ee16e3c960dfa7403dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dec6701c8404ad8e0095ee16e3c960dfa7403dc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
