)]}'
{"id":"openstack%2Fcinder~997469","triplet_id":"openstack%2Fcinder~master~I8cce8cc53633be716a927809fc41088b3468f22c","project":"openstack/cinder","branch":"master","attention_set":{"38081":{"account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"last_update":"2026-09-29 08:16:42.000000000","reason":"Someone else replied on the change"},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2026-09-29 07:59:53.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-26 09:38:58.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-09-26 09:38:58.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}}},"removed_from_attention_set":{"37940":{"account":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"last_update":"2026-09-26 09:37:04.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2026-07-17 03:02:42.000000000","reason":"\u003cGERRIT_ACCOUNT_38081\u003e replied on the change","reason_account":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-10-02 14:16:28.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}}},"hashtags":[],"change_id":"I8cce8cc53633be716a927809fc41088b3468f22c","subject":"[Hitachi] Add support for NVMe-TCP.","status":"NEW","created":"2026-07-16 04:10:09.000000000","updated":"2026-10-02 14:16:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":92,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1216093ad6694ffb5ed6edee48de59a85a04fb49","_number":997469,"virtual_id_number":997469,"owner":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-29 12:44:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-10-02 14:16:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-16 04:50:31.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2026-07-16 04:55:03.000000000","updated_by":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"reviewer":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"state":"CC"},{"updated":"2026-07-16 04:55:03.000000000","updated_by":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"reviewer":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"state":"CC"},{"updated":"2026-07-16 04:55:17.000000000","updated_by":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"reviewer":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"state":"CC"},{"updated":"2026-07-16 05:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 05:14:02.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-07-16 05:25:06.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-07-16 06:00:04.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 06:30:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-16 07:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-16 19:44:06.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2026-07-16 23:31:03.000000000","updated_by":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"reviewer":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"state":"CC"},{"updated":"2026-07-17 03:02:42.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REMOVED"},{"updated":"2026-07-17 11:34:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-24 00:57:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2026-08-01 23:28:58.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-21 15:16:36.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2026-08-26 17:22:28.000000000","updated_by":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"reviewer":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"state":"CC"},{"updated":"2026-09-17 03:41:50.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-09-24 05:44:49.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2026-09-26 09:37:08.000000000","updated_by":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"reviewer":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"state":"CC"},{"updated":"2026-09-26 09:38:58.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-26 09:38:58.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-09-26 10:33:04.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"1cc7dbb08899328919913a5ce433324b41e1ff38","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-16 04:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e2a718440bd37e54a768db096e2535f14c4298c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-16 04:10:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fc62b9fed008c8ff8e22d86fc37cda0db219385b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-16 04:45:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b6595500cd95d668470e11c5b437b82a1b20208a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-16 04:47:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"0cf3b1abbf784bd4c86a95f8173422b77956a4d8","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-16 04:55:03.000000000","message":"Patch Set 3:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/622 : FAILURE in 0h 8m 52s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":3},{"id":"f0eb62503fc1a214a1f73a8f6bb27f3226859c77","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-16 04:55:03.000000000","message":"Patch Set 3:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/885 : FAILURE in 0h 8m 48s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":3},{"id":"4e8043b64e48e748236ab69abc5f06e3c7681f41","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-16 04:55:17.000000000","message":"Patch Set 3:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/530 : FAILURE in 0h 9m 5s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":3},{"id":"de7fe724ac02f2320bb4b7ffc65b544ddc2f14f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 05:08:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/015dc48074b64dcd88f3101623f79e33\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e340e1223d8e4ea6a1bcbb54c4aa5467 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b8bc9039f7cb4ca3ad9a5a77181a7285 : FAILURE in 7m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5f341a09004e459e96d9c2c87351de94 : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c9ef25da59c68801b3d87b74ef7b2c1b11878aa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 05:14:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4806/.","accounts_in_message":[],"_revision_number":1},{"id":"adcc3a328120961b1bc3fde41ce137422b881cad","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 05:25:06.000000000","message":"Patch Set 3:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/cbc5d5c4f2e34a9eb52f15a56cc1e99a/","accounts_in_message":[],"_revision_number":3},{"id":"b2cb53ebdf9f6d288b71c564e385314ba63711ba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-16 05:53:06.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/cbc5d5c4f2e34a9eb52f15a56cc1e99a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5d3c833147eb4071af62ebac5bc68110 : FAILURE in 41m 42s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d0d678ae852a4be8a06bbae813386727 : SUCCESS in 34m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba543eff1ab043499471d112e754815f : SUCCESS in 38m 21s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a4fac69fbbd54a34b4e942021f52384b : FAILURE in 32m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4f84818c3107ac584114e089086a751119043182","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 06:00:04.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":1},{"id":"8f198e70436c3e697d331165dabe2e5e12a16be2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 06:00:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":3},{"id":"66d12078553bf0a0d5fd9d9e64e0ed44fbce53bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-16 06:02:56.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":2},{"id":"67932e24fb456613e29ee88c136534221d616a9f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 06:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4807/.","accounts_in_message":[],"_revision_number":2},{"id":"89a513710337fe7bafc70f32b61604248e600942","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-16 06:30:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5fb44f7cff144ba2a270c10cf69ca553 : SUCCESS in 49m 16s\n- cinder-storpool-tempest-iscsi-multipath build/95da83e6d2d54fc9b8a9c14adbb81da0 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":3},{"id":"6de2a326095887f8090201733df6fec5c51a01f0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-16 07:24:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4808/.","accounts_in_message":[],"_revision_number":3},{"id":"c46537ef84687290edacd721976ddedf8b56bad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 07:49:21.000000000","message":"Patch Set 3: Verified-1\n\n(26 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1888da6559684b349a8025087980024e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be12c327d70846f4b46b85d8f403c5aa : FAILURE in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f1781be5aaf417aa58b177e1e7cad40 : FAILURE in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4284476709f4a32be9dd0b08b271ba2 : FAILURE in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/370fd885bf5541e6b8d03b73a6625e35 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7a7f091c214a37ac84daf812cf2322 : FAILURE in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/80421c5f2fa446ab8f40daa6a75f1e29 : SUCCESS in 1h 03m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9396da4a6fc48b78292fd0c4b19eec0 : SUCCESS in 36m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4159ab19804143f09d2c059d2f346095 : SUCCESS in 1h 03m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/417f243be9724909b1f1c88ea7292767 : SUCCESS in 33m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a98fce0edd444afb3e4dd7dffe012d4 : SUCCESS in 6m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/178d53c9b10d4c67b49710144c41090d : FAILURE in 3m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2728acf63cb042a49b669edf947c2418 : SUCCESS in 4m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69da9eb822d3482da178fe21c732d0c4 : SUCCESS in 4m 12s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/915593bb1c28428fa7c6d1ac00e75bfe : FAILURE in 2m 59s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d434a87a81834c268195b302c283c17b : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6ec461af8b234124aa38653e5d298871 : SUCCESS in 5m 40s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/a9bd174df3154815be4880ab551313ef : SUCCESS in 8m 58s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/44963fab15e248cf8ab9a6ecc681ce18 : FAILURE in 25m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/071b3ffd446d458aa97407e99bf0b99d : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd0f8c140f77449f99166f3a61153873 : SUCCESS in 1h 10m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e0f7b9de307344c3a2788a0d5b0a1abd : FAILURE in 18m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6cebda6f164248ffa81eb7c7e47c8a66 : SUCCESS in 1h 37m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/26d15e319c874447a9a1dc4ae335fd69 : FAILURE in 2h 58m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/40e93c78b57a41e5b687dfd35a2bf30c : SUCCESS in 1h 32m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e0420879b2854465a86fb0b9c84de934 : SUCCESS in 36m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/582500322fb54773baf4e4f6b4b61dbe : SUCCESS in 1h 17m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e90d785ab6f24fceb33c2d318cf05b26 : SUCCESS in 1h 08m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/95a2e3aa183f4ee196abd226c0dbb898 : SUCCESS in 30m 20s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/400a1aec78bf4931933fb2f3c6c43609 : FAILURE in 1h 31m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce872c98fb34f0fb32f9faec5a8cdec : SUCCESS in 25m 57s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"bd095e3d3828f299eca8042334453fd65468b282","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-16 19:44:06.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-997469-2/FC/ : FAILURE in 15h 32m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-997469-2/iSCSI/ : SUCCESS in 2h 25m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"c52d7e0bc90f90a48c5bc5af607b67325133fa2e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2026-07-16 20:10:01.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-997469-1/FC/ : FAILURE in 15h 59m 1s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-69-997469-1/iSCSI/ : UNSTABLE in 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b98f48f11172a52e1656917a9a92ee55ad871613","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-16 23:31:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-2-fc : SUCCESS in 1h 8m 54s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-2-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8e9b9ecd824b9c997c50a0362cb7dad3ac277234","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-17 02:26:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-1-fc : SUCCESS in 2h 16s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-1-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2523b8b9fd938c9982ae794b3629b1ca6a1454d6","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 03:02:42.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_27623\u003e.","accounts_in_message":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"_revision_number":3},{"id":"5eda63932e369b29bfd3e14ba9f7a5897f33c90a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 05:07:12.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"80f60436434a5e402fdbc9cb96d9888185d8efaa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 05:15:51.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4817/.","accounts_in_message":[],"_revision_number":4},{"id":"b01c503db4b96349e57df55a02109a63542e7ae4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 05:20:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a1595e8f484a7482d96105fbaeaa17\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e51b33c1fcb44c99f5511584604b449 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3676d7a02b47491ca014e5ed22163853 : FAILURE in 7m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6ff60d11c27a43ccabdb4dcd4f5e7904 : FAILURE in 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6c1d9bff961613cad4ded02ab175529300867712","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 05:25:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f35443dcc41c902b740d065a3d4a9b2145a8e3e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-17 05:26:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9a7855f34cc04d559f1455ed0e9ec2eb : RETRY_LIMIT in 16m 26s\n- cinder-storpool-tempest-iscsi-multipath build/807d259908644668acd9addd064eb9f7 : RETRY_LIMIT in 17m 30s","accounts_in_message":[],"_revision_number":4},{"id":"45cf0bb13247881c2ea60410e4e144a647cd9513","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 05:47:51.000000000","message":"Patch Set 4:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/db778bd962fb4dd5a32b29d1bf086a06/","accounts_in_message":[],"_revision_number":4},{"id":"e80871200e22ff1b36e6a1ccb46ae5c8c0cafd41","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 05:53:22.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/db778bd962fb4dd5a32b29d1bf086a06\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5ec958478224403892185b8c0e954cf : SUCCESS in 40m 09s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f3d72c44ddb4361937aa4f5f2639863 : SUCCESS in 41m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81e5599d8f3c4f7685a838dc30f45326 : FAILURE in 45m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5099f96079f741c0a75b9a382e976a29 : FAILURE in 43m 41s","accounts_in_message":[],"_revision_number":4},{"id":"4a89713acb8176f7815694d68b8523c6b98efee3","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-17 05:53:49.000000000","message":"Patch Set 4:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/537 : SUCCESS in 0h 46m 16s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":4},{"id":"1f67cd982647b1ea4edb9d8cff2039380e082b16","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-17 05:54:04.000000000","message":"Patch Set 4:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/629 : SUCCESS in 0h 46m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":4},{"id":"1c09fcc31436e2db5254dd15d379a6ba44e250c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 05:59:23.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c54ac33e1d9ccdcfe1449adbaa93d47f2c6c4758","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 06:00:06.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":4},{"id":"1585e234174c85f5414ab046ee8639cacc5337c9","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-17 06:11:16.000000000","message":"Patch Set 5:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/894 : FAILURE in 0h 2m 21s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":5},{"id":"0fafd4154e4e35c4bc4bdd1be84535421fefef72","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 06:13:30.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4819/.","accounts_in_message":[],"_revision_number":5},{"id":"8153f67aacd095a5605763cc4fedd5a812e1a0ed","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-17 06:13:50.000000000","message":"Patch Set 5:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/631 : FAILURE in 0h 11m 35s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":5},{"id":"3e96d87b22de3f8c6089c6e9e200f71a2486759c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-17 06:18:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest build/969ceee85cf7450db38dc5078a19fc2e : RETRY_LIMIT in 17m 13s\n- cinder-storpool-tempest-iscsi-multipath build/88e3578a27f1464a8ef4e651986c1e0e : RETRY_LIMIT in 18m 12s","accounts_in_message":[],"_revision_number":5},{"id":"77ea21209f8a6daea248811e09ca3e0e97b0b5ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 06:21:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d3604b491b4afeb38d4a66751d2b24\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8725a5177ce9404682cb4e52b2c15009 : SUCCESS in 15m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0d201888bed45ffa871c18e9a2ba6dc : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/343eae33c0694f08aff494a986cabf5f : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f26bdf39ed30ce54fb001a9e9a7c8e01708a63d","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-17 06:35:58.000000000","message":"Patch Set 5:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/539 : FAILURE in 0h 33m 39s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":5},{"id":"ce13c54b34a5d32ed7aae9f300fb1b5575927d7d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 06:40:41.000000000","message":"Patch Set 5:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d1dafc31f0c84e8b8aae6c535e4b48fa/","accounts_in_message":[],"_revision_number":5},{"id":"334bf77458ee1eb3558e9710f66e9aa09f3dd192","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 06:53:52.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":5},{"id":"2ea0e234fd6f662031f85329f2ee9a973fab0f59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 07:13:54.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d1dafc31f0c84e8b8aae6c535e4b48fa\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d76ef8c888934ed7a85f80b29f57c4d3 : SUCCESS in 45m 49s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f984de58f9ed42a2b28e1ffa618ecadd : SUCCESS in 36m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4998b05ec70e4117942b88db9ccf1c71 : FAILURE in 40m 47s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d18a785f9d744fcfa18f0217c63935c6 : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":5},{"id":"77762cda99020843914e26f73fce7ec67f58d69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 08:50:20.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3252c84566574b4782c2c2dc5f57f0e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ea2c51420143819fb6c4900ce4f89c : FAILURE in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/169f4e2f94844d949adf5c0c2db17ee6 : SUCCESS in 13m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80bd0509a27e4b89933fcea8acb054d8 : SUCCESS in 8m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2ff84f84b3e46ad8afdef7b482eb22f : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53bc241691c94d1297d7b358154b6bf4 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/955a04fad14a46508a98fc3bca9b3599 : SUCCESS in 1h 04m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0556598576649a4a1292db90e3fbf99 : SUCCESS in 1h 02m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fc77f8731574452bce7b3021cf33869 : SUCCESS in 1h 29m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e1532a3302c4987995b51fb042167db : SUCCESS in 32m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38add6e49184555b9063c8c9f6e3cba : SUCCESS in 10m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21bd317f6003408fbe7aa6e88739b909 : TIMED_OUT in 41m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/60b363c34e5747bc9f99c2d93dd367e5 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1be71403f0b44b04820ddb6cac2fd5f0 : SUCCESS in 6m 47s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/b224d217200e4b79b383f8c09745077d : FAILURE in 18m 13s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8a87d61fd2e04b4788b3598ac117456b : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3ad3561b4713471b9d65410f2d6e8815 : SUCCESS in 5m 56s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/90ec8a6ca37642efa26b1b413915c166 : SUCCESS in 8m 03s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/33409239affe414188ce061c1ece13f0 : FAILURE in 57m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37aabd2cfeec4f8fa91fc3d933159899 : SUCCESS in 6m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d7770a83591a488c804daa80538959d1 : SUCCESS in 44m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2ce38ba491ce457699228220db0c7461 : FAILURE in 2h 20m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0573b3c6602548edbf92839faeef5c63 : SUCCESS in 2h 11m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d02f301548c4485fb1073f140a8017c2 : FAILURE in 2h 45m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/89b999bf7f33406a95e2f191cee84d7e : SUCCESS in 1h 26m 01s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/0326ab845aff424a9328be5f6414a917 : SUCCESS in 37m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3143880a96844291a886e2b3d5cf05db : SUCCESS in 52m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6b39955535c746868ce5a5c7d048be39 : SUCCESS in 1h 08m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/535642d7ec2d4462962a0401a933daff : SUCCESS in 57m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5329cf7855c74a2daee76709fa848282 : SUCCESS in 1h 25m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb923870b6334a25b73136595e03ad17 : SUCCESS in 57m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"f95ab37416e42fc0e8b49f23d28abdf77eddcce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-17 08:58:32.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"400cb892fe6b32193814fc908c0e81148dc5c26c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-17 09:22:24.000000000","message":"Patch Set 6:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4823/.","accounts_in_message":[],"_revision_number":6},{"id":"dca569b9348e87afcd0e72e381ee7c3631be59ca","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-17 09:26:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest build/9a7046c83c4e4f0685e3a423c6aef9e6 : RETRY_LIMIT in 16m 24s\n- cinder-storpool-tempest-iscsi-multipath build/11f49217a0ea442881ef56e469fad300 : RETRY_LIMIT in 16m 45s","accounts_in_message":[],"_revision_number":6},{"id":"6361ac89d83ecbfc9d6f279978925a48da25b2f9","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-17 09:44:38.000000000","message":"Patch Set 6:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/634 : SUCCESS in 0h 45m 47s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":6},{"id":"c4911537257a289fde3f0ffd85fa02b3e1ffbf8b","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-17 09:44:57.000000000","message":"Patch Set 6:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/542 : SUCCESS in 0h 46m 4s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":6},{"id":"188d32b175005489fe979b7b77525bc984ed459c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 09:45:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b655b539bc3c4fa0a53961934480fe22\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5ad4829ae454819aa015d3042cf1a20 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/35f8348b4d4f49f3b133f98a52db4381 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/08b69b8bedfd4eee8aca8738d6e08a43 : SUCCESS in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"068ae484b17d5d996b387f370ad56ecff9242d9d","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-17 09:45:34.000000000","message":"Patch Set 6:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/897 : SUCCESS in 0h 46m 39s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":6},{"id":"365a59ffebf6143a1b6b92ad340b34f4541ed778","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-17 09:52:37.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":6},{"id":"ccdfc6fa3f2d825df624d535da0dcbabb10a94d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 10:07:54.000000000","message":"Patch Set 6:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b9afe430e9e7408b9b4456cc91c1eca4/","accounts_in_message":[],"_revision_number":6},{"id":"2d53a28eef2ab9358be6dec9b13b602288307fbe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-17 10:57:47.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b9afe430e9e7408b9b4456cc91c1eca4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4889e07cbcf348df82e008618843c8ea : SUCCESS in 52m 15s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/719c7db21d8745188e6cdb80cce34c3a : SUCCESS in 43m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca16987e0c264e6b9a4a1ada543385b8 : FAILURE in 1h 58m 44s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0c685e2ae7854b348c1c1f307f20707d : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":6},{"id":"f2dccde09c2c80036caa96db7e143f25bd25461f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 11:33:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb04795c3874b18984a60f38b664801\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754e007fbb9f48989389aceecc738c5f : SUCCESS in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d542bf73bd6c42be9fa8b9b352669187 : SUCCESS in 8m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19c22e726768474a87fa832c9ee0802e : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43ea964022eb45b6b6beb898826ea6bc : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2758213293144d68bf325fbf8e296b4 : SUCCESS in 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/63f38e08d65c43f9881cdec4aece7745 : SUCCESS in 45m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da8c24c4989445738a982ddaf079d856 : SUCCESS in 41m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/764e045e9b564655b952353a289fb5e4 : SUCCESS in 1h 08m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4857905370946f2a1368bf012d8aefc : SUCCESS in 57m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87c2ec95aec84e02a6befaa37060a4cf : SUCCESS in 8m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/414f76e17380499e933c148caf7c3eea : SUCCESS in 16m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/18d545db13434833a1c905cf79bcbb8f : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6a2e3eaad8cc47b7bf3b21ee0124f0c5 : SUCCESS in 5m 18s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/c75ee36bbd9d4ba19652c28bc72fb332 : FAILURE in 14m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/45bf0b5f36b4430ca9add525c4ec4e6e : SUCCESS in 7m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3b6f226c785b49fabeee556702d82e6e : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/1cc20c8afe144511b04d91b53c1b44cf : SUCCESS in 9m 16s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15d96aa1714547b8835b42676a5d9346 : FAILURE in 48m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de89091f7e64412ca02285fc542dfda4 : SUCCESS in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ebbb266fdf94a7fae5682aab5adf36d : SUCCESS in 1h 42m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/03fce4e646cc44d882ebe0d6cc7708cb : FAILURE in 1h 10m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/035aee65e6e74d6aaefb2c7bc8fd7efe : SUCCESS in 1h 46m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/9c615bdbabe94e24af850f9b47a795b6 : FAILURE in 2h 29m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3d4fecade2c54a02b70755a1d2580c96 : SUCCESS in 1h 57m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/def13948b88a45039cda470a5cf26bc4 : SUCCESS in 35m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5876f99e49148459af99d46681e3adf : SUCCESS in 1h 24m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/634a0f5f708b439fa14cd2463a2dec06 : SUCCESS in 1h 34m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1c505db8a6944a85aa3f117890eb3d65 : SUCCESS in 33m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2e88e060cd94ae7a7180f67095c4c7c : SUCCESS in 59m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/019d044ad5eb41918297540d0d96dd05 : SUCCESS in 54m 50s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"88f1ae8f7c3c9a4c7f52ea4874ab97864b12c296","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-17 11:44:11.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_07_2026_11_37_24\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"bfc0f6d39d7b8f086a26cb61c90d0157b99f084d","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-17 14:19:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-6-fc : FAILURE in 10m 35s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-6-iscsi : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":6},{"id":"816ded4f6f4eeb50796128f6176f8e063d49e424","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 07:44:10.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4630/.","accounts_in_message":[],"_revision_number":1},{"id":"fbe7cb1a5e0a831bcae3c8a5dc202456da38a83f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 07:52:32.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4631/.","accounts_in_message":[],"_revision_number":2},{"id":"1866b0b6a498899b8f229a694d6ca485773787b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 09:51:56.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4641/.","accounts_in_message":[],"_revision_number":4},{"id":"a64821a01441575dddd289577c18269c563c98e6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 10:10:40.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4643/.","accounts_in_message":[],"_revision_number":5},{"id":"9cf40361cc8861fff98bf326d7f433bc802ff8b0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-18 10:59:14.000000000","message":"Patch Set 6:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4647/.","accounts_in_message":[],"_revision_number":6},{"id":"87c074dde8f59604402db15e1a9bf7330c865fdb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-24 00:18:38.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/hitachi/hbsd_utils.py\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"5917d459d5783f2ee1d01aed42f20f75802571f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-24 00:20:19.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"6a0ac00fbff41740528f04770709b5beed1ca45f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-24 00:20:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"9b9e4fd1bf18d1cad862562a3141170ad82b732d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 00:27:36.000000000","message":"Patch Set 7:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4732/.","accounts_in_message":[],"_revision_number":7},{"id":"b02b7226d09217f389e5ddf6182895969373ed3d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 00:37:47.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4733/.","accounts_in_message":[],"_revision_number":8},{"id":"fd14ff2f3d5d9500c8b7fb84b1aa43f35f614ccd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-24 00:57:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/997469 : FAILURE in 31m 19s","accounts_in_message":[],"_revision_number":8},{"id":"9dcad6a96487cbcf89089c6787618a7525a1bbbd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 01:02:41.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/98231ca2854a4092b062680d5fee8789/","accounts_in_message":[],"_revision_number":8},{"id":"aa85f33bbbd895cfc66dc35be47bc2afb3db57e0","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-24 01:12:54.000000000","message":"Patch Set 8:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_NVME_CI build/613 : FAILURE in 0h 50m 34s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":8},{"id":"75be4bc3342bc1e76e814fc66d2477f8263cc194","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-24 01:12:55.000000000","message":"Patch Set 8:\n\nHPE AlletraMP FC CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_FC_CI build/705 : FAILURE in 0h 50m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":8},{"id":"0323b002eb05c94586b998871451906efc8ab428","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 01:13:43.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":7},{"id":"58333f855455544495833a6e193c30c8cf20abe8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 01:16:47.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":8},{"id":"26b4a0b5309ba7b4238135f6ba951758ee29ccfd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-24 01:16:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest build/86f7ccd3dc6c427ea6831cbb17614f06 : SUCCESS in 47m 27s\n- cinder-storpool-tempest-iscsi-multipath build/c62604a516ba4ceb9c2ad1613a03e0fe : FAILURE in 55m 06s","accounts_in_message":[],"_revision_number":8},{"id":"8429bab829ae87bf6a4a3164d7d799d735686804","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 01:21:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025ab611ac294718afb3b35abe5c6633\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2f0eff27a69426fa3826e37a2193a79 : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a306f70a01fb49339e5521a6825d63e5 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/25a960fa2f23445fae2d8f86b09151f2 : TIMED_OUT in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1ff89fc7eb10aae24fcb077c295f9b22a3c23011","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-24 01:22:59.000000000","message":"Patch Set 8:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/971 : FAILURE in 1h 0m 40s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":8},{"id":"c94387e8b21979b40098f32846eed54327f5f00b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 01:53:22.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/98231ca2854a4092b062680d5fee8789\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2416b974f1d14a7ea459fadbaf2d36a7 : SUCCESS in 41m 10s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4ecac8bbb8334fcdb8b47aea2cfd7d9a : SUCCESS in 37m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/002a9d81792f44d8b756816a373f46e4 : SUCCESS in 47m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb7143212a7d4e35af22064d3b5d5eac : NODE_FAILURE Node request 200-0000520564 failed in 0s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/823f99800e714d8daa719ebbb2265426 : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"deeff99e6dd337009ea44e157efc57433c5411ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:56:04.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aef118fc78b42ebbd9ce7a8469cd17e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4553a97e7da640afbee47b635885af38 : FAILURE in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3833919d5055453bb7ef7a5f2e90e444 : SUCCESS in 8m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8ca27ce73f44d41b47a1ffd2376968c : SUCCESS in 7m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c951257a8f345e8a5d2529bd267d2a9 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7325c339ff4b19a5dfce70b15f7f14 : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/f40d95445c914f5ab7b41ba5150929f8 : SUCCESS in 1h 09m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ffe16f0a47243038a2b950ae56ce462 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/726d52fe7f2a4d97ac8936ffccafb47c : SUCCESS in 1h 27m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb1ad39dd2204ee885faa2a8ad94d941 : SUCCESS in 31m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d89f7052b92841e58bbe0030718627d7 : SUCCESS in 5m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3727cd319424bddad204090d7861f8a : TIMED_OUT in 40m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b89d934f71a4de2be87518682340c97 : SUCCESS in 4m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7914ff0a8c5c4a2498f7b60d52081031 : SUCCESS in 6m 55s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/8b4210418979449ca20eeac81e256afb : FAILURE in 8m 26s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a02d80b5f5854809b47af0e8391afaf4 : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8ed53cf7694f448db06297134e53d833 : SUCCESS in 5m 29s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8510a72643e5485fab849059fa499ae3 : SUCCESS in 8m 54s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9466e2d6d84f47b9a712c21b87f3f271 : FAILURE in 55m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e349aea5e2074a6db03fde638c1e83b1 : SUCCESS in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c9f6f38b858247bdb2768b339d3d9a7d : SUCCESS in 1h 28m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/91bc43c7c84f4fc4acd7f72b03f93a1c : FAILURE in 59m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ad61553ff2e41cea6a726262d473804 : SUCCESS in 1h 21m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/1b2267899b9d43fca1f3ad05819b886e : FAILURE in 2h 29m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c9c1c125c54541c0af420b41d46713b0 : FAILURE in 1h 21m 51s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/429a8ede92e04f6686204d2583fbff5f : SUCCESS in 35m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/21aab95fc77b4fe19367da6db3f1955d : SUCCESS in 1h 10m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d0c03dc451ed48ae80f17d8acf40471c : SUCCESS in 1h 11m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/080ea12fd37a489780b8ee2b26eed6a0 : SUCCESS in 30m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3e38949deea46e9b71a57e8b50cba82 : FAILURE in 1h 29m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d9018069b1244918abc1304c8cb0d1f : SUCCESS in 28m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"8f762cb5f36eb24b9990aae45fb130edcc29eebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-24 04:57:39.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"4dbf3b693a400d1d9724facdf30e9beb1b5fd8d3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 05:06:20.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4735/.","accounts_in_message":[],"_revision_number":9},{"id":"740673842bb3fbfceca920875f8923db8ef27031","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-24 05:08:09.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4911/.","accounts_in_message":[],"_revision_number":9},{"id":"2825dd4f2ee771080940c27a0f50d90017081a7f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 05:17:41.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/e61555598a1f4bb28c7caad98bd3f5e4/","accounts_in_message":[],"_revision_number":9},{"id":"1eaf05d02c5220028e10afec59624c991a5fda40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 05:21:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d2246d7c5242cbbcf589cbcee28c69\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16e728872e3549c4836a16bd0c4a2d01 : SUCCESS in 16m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5bfb07cf0f11410081965ec92836f85d : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/efafaea633064a588a60248542a8cc88 : SUCCESS in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"38842f7bdaeb1020888d1aa68d279a4020421dc2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 05:52:21.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":9},{"id":"1736336fe4ae9f75920c76ba64ef9d0de5329955","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-24 05:53:34.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e61555598a1f4bb28c7caad98bd3f5e4\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3eccfefa0ea14cae9a949c7c80e0d24a : SUCCESS in 44m 18s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/093d37189cb24d6e97c490dced33fb1e : SUCCESS in 35m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc3613afc17641b69561f54809b4cff3 : SUCCESS in 55m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e152fa3323bd44cf8bc369fe2c0ee19c : SUCCESS in 37m 24s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ef327c7b3511492dae2f67828d66d66d : FAILURE in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9f99f4ae8e7791b1020af25d908c4238437b8466","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-24 06:08:01.000000000","message":"Patch Set 9:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/973 : FAILURE in 0h 27m 7s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":9},{"id":"b900662ac8238e16aa140002950366d68bf6dbe6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2026-07-24 06:24:14.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/997469 : FAILURE in 30m 06s","accounts_in_message":[],"_revision_number":9},{"id":"9f92d505d3ec89eb3288885ca77414cc539a3c75","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-24 07:08:31.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/59b2a4aae022458b8602cbf109300544 : SUCCESS in 44m 05s\n- cinder-storpool-tempest-iscsi-multipath build/5de5b2acaa38411c965ecd6d10c1f01e : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":9},{"id":"8c5e8ec4eaa390815bd9088dca578691471d7c1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 08:06:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b03687355644906b368488c53b4958b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16a23ed2187749279c232a4622c83e85 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b187354420904c60a7370af17d8ccd29 : SUCCESS in 10m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea4b52a79f7f4577803279b61650c8ba : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de091afebd354310bf61fd982063e3b2 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7446571ef4c548dea12b73ed1ef59297 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/adee416b156f40a3b29ee1412893b04b : SUCCESS in 1h 05m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d6fd6c99c6e4630975f17bb9f329b69 : SUCCESS in 30m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db41c081864b4b098bab73f3f70c33fc : SUCCESS in 1h 23m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42190d7802ac43faae792ac456212a29 : SUCCESS in 49m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a3471e4deb41698f549bbef2f7af6b : SUCCESS in 9m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/40277ca7fe7a4fc7b808f2ded8c02491 : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fb074bd33324484aaec1c6e924922acd : SUCCESS in 4m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d90afdfef3174778991970ede75ea90b : SUCCESS in 3m 53s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/a2e9b8bc007147ed873ffde636e6301c : FAILURE in 10m 10s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dc79d11ae12c4798b0ad071c5ab2fca2 : SUCCESS in 8m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6d26594ea634ec688635dcbc36701d7 : SUCCESS in 5m 51s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/8fbfe439211242cdb23af66b96785b3d : SUCCESS in 9m 27s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef2edff52cdf4fd3848298aaf6a3af3e : FAILURE in 38m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e7990a524794ea283339e7ca7ccbd9d : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3665435e4010417baa2d79a1779dc657 : SUCCESS in 1h 39m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/046144c264d64c258cf227a9edced0d3 : FAILURE in 1h 05m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b77afc45a83c46da93010d2e587d014f : SUCCESS in 1h 51m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/74c0b0f964d948bb83f6c42f21ed845b : TIMED_OUT in 3h 07m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/00cf1423e4874a28b6809e7de4dd9813 : FAILURE in 1h 27m 03s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c43abbd0cf2a42488c50e51afc8c5d2c : SUCCESS in 25m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/49e054ecaacc470e868485057e7914e8 : SUCCESS in 1h 17m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/66bb74bd398b4356a3be4d5c5a71d51e : SUCCESS in 1h 17m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9e82c22d0d0349778359d3d82cc2c4a8 : SUCCESS in 24m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38b7dee876914a07a29dc46f5138bb56 : SUCCESS in 1h 47m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88e03e37bbfe496093e577b77a2ff646 : SUCCESS in 55m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":9},{"id":"b73864319a40481ebe47618f2c5dfd38ebf1ab6a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-24 08:17:24.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_07_2026_08_10_21\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"c24eaa216bbf01e4345b78c47525ee965eebfefb","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-24 12:54:53.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-9-fc : FAILURE in 10m 25s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-9-iscsi : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":9},{"id":"cb941f66f66fe5740ad728e7b58a8baccca8271e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-07-24 23:02:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-9-fc : SUCCESS in 1h 35s\n- hitachi_ci_nvmetcp https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-9-nvmetcp : SUCCESS in 1h 4m 22s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-9-iscsi : SUCCESS in 1h 1m 52s","accounts_in_message":[],"_revision_number":9},{"id":"2789b473f81dc16df775593386b724a152126b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-29 12:36:24.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"82b02063be8421e16dc60f9e8969f382f69db690","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-29 12:40:56.000000000","message":"Patch Set 10:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1009 : FAILURE in 0h 2m 21s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":10},{"id":"78c9263c6b59a9a9726166dd1630de4c66d932b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-29 12:56:46.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4948/.","accounts_in_message":[],"_revision_number":10},{"id":"cd90b5f4973350fda3674af0a46b8f51a251812a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-29 13:00:01.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4772/.","accounts_in_message":[],"_revision_number":10},{"id":"bcb3149225e056a7ce4666b419b38f2ca846feb8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-29 13:14:54.000000000","message":"Patch Set 10:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/dc2306782e174f06b7eff71ceb7cfd81/","accounts_in_message":[],"_revision_number":10},{"id":"566da3fb6150f3a86fce1b5752158724c607255f","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-29 13:23:19.000000000","message":"Patch Set 10:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/652 : SUCCESS in 0h 46m 31s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":10},{"id":"355c7423e40ce5f29439525bcd9ee86a597d7823","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-29 13:28:08.000000000","message":"Patch Set 10:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/744 : SUCCESS in 0h 47m 9s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":10},{"id":"a1ac8791180442e9dd00937887fe4ae672725cc4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-29 13:29:00.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest build/e3192173b8274f4789c422ceb866d037 : RETRY_LIMIT in 10m 55s\n- cinder-storpool-tempest-iscsi-multipath build/d47d3f245ed1421596c2c732d606b9ef : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":10},{"id":"7a96bd6e237f8208886638c8b60722baa485d879","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-29 13:31:32.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":10},{"id":"890d294914c6615aadce2f33fb1d41cf922cae54","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-29 14:31:08.000000000","message":"Patch Set 10:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dc2306782e174f06b7eff71ceb7cfd81\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/12189bb2a94d495fb5cc0dca9e17227d : SUCCESS in 40m 07s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d687456dfa04b0892fadd954204f254 : SUCCESS in 35m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1fb9e1932f164e62b1186448ac35b1a6 : SUCCESS in 50m 02s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f51b9ed22f1043c79d06fb4c49b74863 : SUCCESS in 42m 04s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a2d28008eacd4830a5f59a69e6d43267 : FAILURE in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7864eed0dbc2ae85eb1790be82de0dc270772169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 14:59:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e4c878170b460ab6d9c244ebd543a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282e7bcd67ef4019a9b178e93e1acf6a : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73592366f5c0461fbf731bc54f307092 : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/625c603bac6644638e463bddd15cfb37 : SUCCESS in 10m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7576eb943c24d32887d6803a32e87fb : TIMED_OUT in 41m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f086490d6c954ab088e457fb8f7c4e9d : SUCCESS in 14m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/46a2676cb2754647b2d0507c9e78fd21 : SUCCESS in 48m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a3a13adf24b421783c5ddccf20dd587 : SUCCESS in 1h 17m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fcc8d4adf314074a323d47248e1d3f6 : SUCCESS in 1h 37m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fefea10094d45c5909492de6dd857bc : SUCCESS in 1h 03m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8582c6d71a244a4b4b8eea65f2092bf : SUCCESS in 8m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/498580e6196a4764862730fac8a48259 : SUCCESS in 25m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/871f00fbbc544ed0a083fbdec5450f63 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0aff799528646a38869d728dfb653e9 : SUCCESS in 6m 41s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/7884e4d39c754fdcbb63a6b3601b51b7 : FAILURE in 8m 52s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e3f9b5f7ce514d3b87507265b2f41b35 : SUCCESS in 6m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ee4577893bf0450a9be998f29b41e874 : SUCCESS in 6m 24s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5fc5181980984613bc2f892429a23e26 : SUCCESS in 7m 33s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a5e8b571fb543fc9326ace52a4b2f08 : FAILURE in 50m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de43e75a4949463095f62ee3b5ee36fd : SUCCESS in 6m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a0beeda251404b80babce21e2b4af830 : SUCCESS in 1h 34m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/881b9061e62d4ff5a376a93a6dfdff14 : SUCCESS in 2h 12m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e79b7dbe04446b4b80aa58d439371f9 : SUCCESS in 1h 37m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/39e30bf02dc14cb3a86e2c824fd5ea2b : FAILURE in 1h 58m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/05b510ca615b4c9b8942f8e03d66221d : FAILURE in 1h 52m 57s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/d78c5553921f42f1b722ff7503d126ac : SUCCESS in 36m 50s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6d7db0feca8f433b8d200f815307cea1 : SUCCESS in 1h 15m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6585f140aa9b4c12b20d37349dfa91b4 : SUCCESS in 1h 31m 35s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/79fbf78736234da1ae0c50c28e227aab : SUCCESS in 45m 34s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa6b72ec0eb34cfa82ba220331161b42 : SUCCESS in 40m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52390857e2504343afff0f116cd854f9 : SUCCESS in 55m 42s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"e61b005d8ac5690f1f0f322261e8e0accdf325e0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-29 15:09:36.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_07_2026_15_02_39\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"f4573bc2e6cdf143719df76693d8629720fecd57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 16:25:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d101d6754dd4b13b7ce1761da2c08a2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71dfd66299584f2da68881a8917c57e1 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b2f8900d8e346a68e997dd442b353ce : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a79656d3ca314bce867d20a1a5c22df1 : SUCCESS in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8ccb8612569e66c8d39d1f24ad82bc18009ab0be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-29 23:29:13.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"e656d234f3bb6b334e27145c7723020282acb307","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-29 23:33:05.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1017 : FAILURE in 0h 3m 32s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":11},{"id":"01a746b5e7d9766e8370badfd0f3f45379680c29","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-29 23:39:02.000000000","message":"Patch Set 11:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4779/.","accounts_in_message":[],"_revision_number":11},{"id":"5e79d750eb0484ea42745ae1ee2cbf54a6995d1c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-29 23:39:50.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4955/.","accounts_in_message":[],"_revision_number":11},{"id":"12c442701f2c23a4ec8964ed7336878f26981b42","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 00:03:56.000000000","message":"Patch Set 11:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b4fb9b7f966c4627b42329618c9375d8/","accounts_in_message":[],"_revision_number":11},{"id":"50029add3ece422c68fa60a403a5253da00b918a","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-30 00:16:17.000000000","message":"Patch Set 11:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/659 : SUCCESS in 0h 46m 46s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":11},{"id":"1a3e5b8500d43ef58ac6e8fcf50486a62d63fd8e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-30 00:16:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b4fb9b7f966c4627b42329618c9375d8\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e60a049e47c34d3db8adeb27e94d21fd : SUCCESS in 38m 02s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b640d6f16dfd438b80049140a91dbff7 : SUCCESS in 34m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/95c932fe836f4737b8d546a350cd8665 : SUCCESS in 42m 35s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbb14c5f1d5847e584de5c6a7d6f51ed : SUCCESS in 36m 39s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dad9b23717024eb7ab58e1c2cbbf45aa : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a063fedc2a4ff6214af2cdfbaa2c6a0e83edf7b9","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-30 00:18:11.000000000","message":"Patch Set 11:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/751 : SUCCESS in 0h 48m 37s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":11},{"id":"991107525ac7a3daaf127837a0d6ac3b7b745ad2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-30 00:21:43.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d5694ca843d84e628ca37bb190f09aeb : SUCCESS in 47m 17s\n- cinder-storpool-tempest-iscsi-multipath build/741643a1647b4f88a61e592c0cb942af : FAILURE in 51m 06s","accounts_in_message":[],"_revision_number":11},{"id":"13995c72104717eaca990abf1f80979c6f64b4ea","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 00:23:05.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":11},{"id":"dc536b4adc3048454a0b751ffa1614a9d58c4f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 02:06:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d5b94e01bae4ae2862a1d207f79d6b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb264ec46bcb4f7db5cf4af6bc52c5ae : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64b65aff9a6346eb8223b95914e5b902 : SUCCESS in 8m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/800894d7937446948b5278789acc2509 : SUCCESS in 7m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b604e6b25ec48cbaf3caa84086ebca9 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15591e0bdbe346e8aefeb297c6fea9c4 : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/68e154449c74443f952af78da8f4f226 : SUCCESS in 1h 18m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b8f9ecce8754b4ba9ffd31fc97069ad : SUCCESS in 1h 03m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59048c78352849ea9a29485f6c66bc6d : SUCCESS in 1h 22m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d86018756fc460abb398a42eb6c0df3 : SUCCESS in 26m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2d37626f00401cb84ef6f52f95fe0a : SUCCESS in 9m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4a2590bd57c4a6a9f670f7c0d118af6 : FAILURE in 25m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/296444fa0b91448b9baad71cf5882a86 : SUCCESS in 4m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/15c2e7f026c242b4b5678056427e8efd : SUCCESS in 6m 12s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/95785fd06b4b439089d32aacc977bd6d : FAILURE in 10m 10s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9e102167d52c48daa696fdb9956981b6 : SUCCESS in 5m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67c4346d73b44b5c8dfcc9b8a622c568 : SUCCESS in 5m 43s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/0f80fdd341414e919c3129a3908d7d18 : SUCCESS in 10m 08s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/54621b48d0b949d9af6a95ac706c23c2 : FAILURE in 24m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/855b2260a4394978b09425b042878d67 : SUCCESS in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/13be9f116c2c4514bbf5e013c13cc74f : SUCCESS in 51m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/30cdaecf66d8469a83b3dc3ded006dc7 : FAILURE in 1h 57m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/14651d14ac49410f9a3476164cc79b76 : SUCCESS in 1h 25m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/27b723a15c494d88854d5549f6964bd7 : FAILURE in 2h 34m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/7e436354a490433fa08a10953a168448 : FAILURE in 1h 50m 59s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c015d60bc2c749eb917159761d38e77e : SUCCESS in 20m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/db7f4ff940f240cd963dc27e5fd227d1 : SUCCESS in 1h 14m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c1d431b38e54a7fb704c4e53562ef88 : SUCCESS in 1h 12m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4aa2d2e02f574d45b0a5d260f2867000 : SUCCESS in 43m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b814947b7d94f0da751a3a7a093faa8 : SUCCESS in 1h 28m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db17b121d3994faca29f6b6c8dd3c970 : SUCCESS in 29m 14s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"4c34c380512dcc690222d5b1b2d6371288148d53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 02:16:50.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_07_2026_02_10_00\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"0b406c0f065be05f7e0ff345ccb2ba21b7a1923a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 04:35:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c8d111258a41a78e3ec411f8d80b3d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7232c78c444d409c840e4c0a079658a9 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48b5d6b50eb94ec98e54f300d58e3caa : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f4e16a60e3ab497683f610ae8f1e4f3e : SUCCESS in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5404e0fd72926c800321132c213f6efda065f14b","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-30 05:22:35.000000000","message":"Patch Set 11:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1018 : FAILURE in 0h 2m 15s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":11},{"id":"5f7191ce78e6f34efc0a3ed994eea8b23c39a203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-31 10:46:15.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"45f0c16f97d15e2d03e709c2aba9dc9437653387","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 11:07:03.000000000","message":"Patch Set 12:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4804/.","accounts_in_message":[],"_revision_number":12},{"id":"2b70e7829360b35bce76ec81395cb29244662e63","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 11:11:23.000000000","message":"Patch Set 12:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4980/.","accounts_in_message":[],"_revision_number":12},{"id":"ed0b7a6f6cf17af747c04b729f1477769a3fbff3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 11:20:32.000000000","message":"Patch Set 12:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/892dd64ba4354e06a60349de9b82abbb/","accounts_in_message":[],"_revision_number":12},{"id":"0f9d384a613c07c95d881dc6b8ffc2c54c559930","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 11:40:42.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":12},{"id":"39461b915941c95273612a6e3be91cc449202d87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:55:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd950d9e27e544d9b95ccd9134911dfa\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e1a094ce86641aa8132e4a0efc5440e : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/980d4724027c4d1bb71efa0728cbb8a9 : SUCCESS in 16m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fa87a8cfa7f547b1b7b50bad195a54d9 : SUCCESS in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"be35d5fb7581f01ee04b4ba988ed341b4b7369a7","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 12:23:29.000000000","message":"Patch Set 12:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/678 : SUCCESS in 0h 48m 46s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":12},{"id":"0c55b1368fc62896be3c69b52ee33f32def8a7ad","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 12:24:39.000000000","message":"Patch Set 12:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1039 : SUCCESS in 0h 49m 18s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":12},{"id":"8714fe42f43e80550d445d34f3c334a740800ec3","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 12:24:53.000000000","message":"Patch Set 12:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/769 : SUCCESS in 0h 48m 47s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":12},{"id":"804d151b0835a06f76255ab39301fdbf21fc4858","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 12:29:15.000000000","message":"Patch Set 12:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/892dd64ba4354e06a60349de9b82abbb\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8d134804ead345a2ae526105206278f8 : SUCCESS in 44m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/21e5e975bc9e40feab76e427a379fc05 : FAILURE in 33m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/114d2b3a75864ea9ad8ab023e79efcd4 : SUCCESS in 42m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e9c1d7e38e5d431e84f282a1a2b84a67 : SUCCESS in 37m 11s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23ff295786c5455e8e5eed2ba2f49a73 : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e4c4bc75d0c683921fc1b9bd972db03a4c07f9c9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-31 12:31:10.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/66f265773b064120be585faa0ce08544 : SUCCESS in 42m 51s\n- cinder-storpool-tempest-iscsi-multipath build/d7db7b0fc9d34ee6a50942483f214188 : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":12},{"id":"338495936af9dd0a751d99166238c9b1d028d9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 14:14:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82bd4cd3edf7461986325fa9bc1fcf88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8609df565814130a1146d5301acd8cd : FAILURE in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e82bcaaca2dd46bcbdb5e3f843c05544 : SUCCESS in 10m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d778b622f2934b859fb94cf0cba30dec : SUCCESS in 8m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b17d5a9978f74ab0b0b71f08f24a23fb : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7333561b7e440d8bd064620c4d3be01 : SUCCESS in 16m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/670a0bd51e79406d86da80f6e0b082ea : SUCCESS in 1h 04m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eaadbc7d0e3e4f1a9661def30a52df80 : SUCCESS in 1h 06m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c8e45287b014257a73c33a35491518e : SUCCESS in 1h 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e58a0785a6164620bf6cbe8417ee7272 : SUCCESS in 58m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b62f06e43814f1ca1e19a3d7079c51e : SUCCESS in 5m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2df2d79be85547b68d2b9d2bebae2ac6 : TIMED_OUT in 40m 59s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e2789484d014d22b9ae24520499683c : SUCCESS in 4m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/844cc97330a34bdf86f3288be712cc0e : SUCCESS in 6m 37s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/2083f938b5764832b64562b2bb2cdd63 : FAILURE in 12m 08s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3ee90f81a6a548caa9ac0c43c751df11 : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b714bc7d739c4827ae325f1e3036e87a : SUCCESS in 6m 14s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9d9b5887721a4d59ae8db7bfdbbe5008 : SUCCESS in 9m 08s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/52110489f228476dbd3b5aee5ea6d3af : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/006f69a328af4f8f82db2797ac6417d4 : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3e1d37f4a2f460caef16c77ffd7cc3c : SUCCESS in 1h 30m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1b61e54d9c0f4511b8104e269f7b2576 : FAILURE in 2h 00m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de16d5c8bee54855bafbfb2dd293f916 : SUCCESS in 1h 39m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/6e9fe084fa714842afab142b87fa848a : TIMED_OUT in 3h 22m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/32db612f434e4cef8559d5bbc4e498e0 : FAILURE in 1h 28m 06s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/7e168cca102a46eaab32551d795d4b9e : SUCCESS in 30m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/11f669d9c6b142c0b8b4d9c6c34cde44 : SUCCESS in 44m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3664b898f3954c48b502b9a8567102c7 : SUCCESS in 1h 37m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/67ae9901d7fd40038e919b446406d85e : SUCCESS in 42m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fde2b36e9b7b44efb30a98d7a6ce3da2 : SUCCESS in 1h 20m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60a2817ec5754560a74e5ea5c3935c9e : SUCCESS in 39m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":12},{"id":"fd822f1fa79ee280766face913bf2f1f41f6df32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-31 14:27:49.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"0ed652c48fa8a7fcbf737f412a971b6a754c4af6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 14:41:20.000000000","message":"Patch Set 13:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4806/.","accounts_in_message":[],"_revision_number":13},{"id":"d0e700216605650df711e1e13aa25a3ebd93397b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-31 14:47:51.000000000","message":"Patch Set 13:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4982/.","accounts_in_message":[],"_revision_number":13},{"id":"5503e1c9e23fd04d218ffb93c4843eface32735c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 14:56:46.000000000","message":"Patch Set 13:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/20760b4b9892464c97652e3fcae9ff68/","accounts_in_message":[],"_revision_number":13},{"id":"c41ff1d53e7219e75b6a45768bf0de9dfbbcb253","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-07-31 15:15:02.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/20760b4b9892464c97652e3fcae9ff68\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6fbb4cd58cbe41cfb08d3bba668fa908 : SUCCESS in 41m 26s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c4723fb82254cbc80c435999ecf1504 : NODE_FAILURE Node request 200-0000628502 failed in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e695a77ba83d4e2ea9908f49e8412ddd : SUCCESS in 46m 41s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ae9670f374084165b37be19cc0b1a022 : SUCCESS in 37m 21s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e2dd84cfbcc047f28f1d7ca49cf8f630 : FAILURE in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"61199de6f3c380734081c620cb599b4cdf31153d","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-07-31 15:17:54.000000000","message":"Patch Set 13:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/771 : SUCCESS in 0h 49m 41s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":13},{"id":"a005b73c2c8040831ab44839535fb1643e18a2cc","author":{"_account_id":39087,"name":"Achintya Mishra","display_name":"Achintya Mishra","email":"achintya.mishra@hpe.com","username":"achintya-mishra"},"date":"2026-07-31 15:18:53.000000000","message":"Patch Set 13:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/680 : SUCCESS in 0h 50m 43s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":13},{"id":"d942f8a3322fafd749f94798fb05a134f54efe4f","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-07-31 15:19:18.000000000","message":"Patch Set 13:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1041 : SUCCESS in 0h 51m 4s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":13},{"id":"ad160b46c6aadf6bfee85e3e1379de34d9f4f836","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 15:31:37.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":13},{"id":"bf81903b6c9dd84ecfe77832583b2c24f1270cc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 15:36:26.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14e162e3c9aa4d44860221333b76c6b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2650a1bf2dc6498eaf4bb52985c4904a : FAILURE in 29m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5093b874fc24f7bb2f3b0e029a90205 : SUCCESS in 17m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d26ebdbd34ba47c3ba248128625f6014 : SUCCESS in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ad73289fc94886dd2f57d776a953a289aa304af7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-31 15:50:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/f265e420d1a0495c9ab15b01d6fc0010 : SUCCESS in 44m 46s\n- cinder-storpool-tempest-iscsi-multipath build/5972101ae3fb479791902db103d0907c : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":13},{"id":"b541cd565c5cbccd39c35efdf346ab4cea3b739d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 17:36:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea9069516cb41bcaf469d747ebe5ccd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1363c82ea9e04694ba963cf4e545fc3a : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a867337588f54a639386c3ae50b5946a : SUCCESS in 11m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b779e0cba8014ac1a6f369d607996a33 : SUCCESS in 9m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ca242434bc24562b11b043f05c53acc : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4b8f61216546ad8afeba8ce29e2ec4 : SUCCESS in 13m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/f85d9f5d23784b5cbe61df42589f64a5 : SUCCESS in 1h 04m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb23b9a803b24ad7bc47abe7680216a9 : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ce4834e8ff4f1e9e19ecf5bfc0ed41 : SUCCESS in 1h 30m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d689e983b4a64e069c5ba57fd740e551 : SUCCESS in 1h 05m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe15b32074242dca48667d1f434dc96 : SUCCESS in 6m 37s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/374b1fd22c574e7cbf9c85fe0569dc32 : TIMED_OUT in 40m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f009a486340a4684b783298cc66baa87 : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4bc86114c0e496ea1d2788f92785e9e : SUCCESS in 8m 24s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/27eaf515a5a84182be890d180a56b711 : FAILURE in 8m 45s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19068c17bf344c6585ded8e3d961f738 : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/10f35961deca45f3ba0082d8f6fff7b0 : SUCCESS in 5m 59s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/61869a9342e24b5dba68c35ddb19a08d : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/553db64693b14a20831a67482c5143f8 : FAILURE in 43m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f61257966d2443c8f8334cdbf7ddca6 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61971a2963d847e5befd863f269ec223 : SUCCESS in 52m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b18a8e83154d4496bf843c5bfd9cd13c : FAILURE in 2h 09m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df4d8fd633a14863a636cd1ca70d8531 : SUCCESS in 1h 30m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/444aabe3c9424bb0b6b05c27ff73d757 : TIMED_OUT in 3h 07m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/0a32df9ebf4a4d73b325ce5d10cd2be1 : FAILURE in 1h 43m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/ccd62ad9fbe340c1a2e29ddd854760b9 : SUCCESS in 16m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6ee5719dd6541d399072a772095ef51 : SUCCESS in 1h 15m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/51119619aca84147aeb1e584211b53fe : SUCCESS in 1h 26m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/296ee87c3ac94fcdae48f12c0aa415cd : SUCCESS in 46m 05s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2970d6a03364c958d93d60e6e32fbf0 : FAILURE in 1h 26m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25145a0cfb8b4a589fc9199b0f6fcea2 : SUCCESS in 58m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"761641a539a4d5c3c08b3ab2ecd9ebbd38552d2b","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-31 19:02:49.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"39cecaede1a09906071fe54434a29637bb754478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 19:27:47.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e8c1725c564d5ba5afa1caeb8590b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/86e5eb606d164f96a17428c0e19717bd : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3ec9c6ae708485b89841facc2de30b7 : SUCCESS in 16m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/86a9f4ae7ac74c5f9300be0404d6e344 : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"21a7c9b15732bd22099250d1336e825618ee379a","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-31 20:22:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a6c6cc78d7b813262c96baca2abbc6a305e2756a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 20:47:12.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5fe69a8d2c475599ff54b012e6c394\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe0863cc2d4b4a9684dfe83859378787 : SUCCESS in 22m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ecfd812e2f5d48c192e757d66ad91d98 : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0372f9bf24e4768be85c832507c18fe : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8cd2875021da2380a1a580a29ff933659e94f693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 21:48:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3e6fc5451246b98a230cb652ae8632\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707ea272342341af827ad9777484f094 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bf66bbfe4144a398833945218e9302e : SUCCESS in 11m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f7a04c3f1e74160bdc8bad7e3d23418 : SUCCESS in 9m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6c2571eb6184a41a0dc1dd1f703780f : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed83dc29b4324a4090c4ff4a53e3b3e0 : SUCCESS in 15m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/af3b6c10ed7a47909b427339487543e4 : SUCCESS in 36m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7ee7c55c12043d49ca0c560794b4b0c : SUCCESS in 1h 04m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50e8515bc8114e0e938046402d0787a3 : SUCCESS in 1h 35m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/568aecd5965c47c88b18359d267c0f1e : SUCCESS in 1h 02m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b88f34d66294cc7a5ffc976411bab66 : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/742d01d2dbf3421b944e207c94fcba13 : TIMED_OUT in 41m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cfbc0cf1378c4fbaa561c523639b288a : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be4059d6b6e64121b0cf09fe9ea84d44 : SUCCESS in 3m 18s (non-voting)\n- cinder-without-eventlet https://zuul.opendev.org/t/openstack/build/1a979daf392a40b98d0b788218ad3568 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d9696ac7d1b44cabb11d7bcb1dc53a0 : SUCCESS in 5m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7e73671e2a534d568fc6c7c42e769be4 : SUCCESS in 6m 03s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/11c3d9730d2e4b9588706d127f794c90 : SUCCESS in 7m 28s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ae4f4e9df2148829fed1cd402cadd6d : SUCCESS in 37m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b545fd690cd24a1ca4188891f3c18d33 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/39aace7eb6224cc2bd1a0b311ed47ff7 : SUCCESS in 1h 37m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a9a72c805d584a9c88db64ecb32f169e : SUCCESS in 1h 12m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/870e0454251a4ece9667245aca49fdda : SUCCESS in 1h 17m 39s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/ea72c3722ccd417ab008fe0e090afd83 : FAILURE in 2h 42m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/16196f3b59f84e2f95e435747ffe5875 : FAILURE in 1h 37m 24s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6fcf78360e3047caa4b8429954ca6b9a : SUCCESS in 21m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ba28687cdce94153996a44f17e917cc2 : SUCCESS in 1h 19m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/abfc39fd70d14caa9992f56cce250c71 : SUCCESS in 1h 32m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7484a79102f4991867401ddb65b1f58 : SUCCESS in 48m 07s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05ef1581c2444f40a92fd21a1b891301 : SUCCESS in 1h 42m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eec6e7bf8c4d4c88bfa4c8ece2202db9 : SUCCESS in 59m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":13},{"id":"7d44edbe7a4fc677ef5710dc689e6270d0f1b755","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 21:58:28.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_07_2026_21_51_33\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"c2b58b8e0bb68bc8fdf43a84e9bde7e412840575","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-07-31 23:48:56.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/9b7899f3a39a497784011abbb7638df3 : SUCCESS in 45m 59s\n- cinder-storpool-tempest-iscsi-multipath build/d80b15729074498d8c2686a32ea99f1a : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":13},{"id":"af171c6e0601b662b559478388838968bee1443e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-08-01 23:28:58.000000000","message":"Patch Set 13:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/997469/13/20260802032716/NEC-ISCSI : FAILURE in 4h 55m 40s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/69/997469/13/20260802032716/NEC-FC : FAILURE in 5h 01m 39s","accounts_in_message":[],"_revision_number":13},{"id":"09481e64c023dd3eaa1c040b6ae33816d60024eb","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-21 15:16:36.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"46e05fee34dc0b298f9c7e8e5a19d0be97b502b6","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-24 20:01:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0fb410ec257e4b6ebdb02ca3e8535c1622d7edb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-26 15:21:46.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"38eaac662302371b124b9ebfd65295db748a1708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-26 15:28:18.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"20bfe8deec717311b9249805b8fb632f1f7f794f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 15:31:37.000000000","message":"Patch Set 14:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5268/.","accounts_in_message":[],"_revision_number":14},{"id":"326d16e0bb81519584bd6b9e78c521962e946c18","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 15:40:57.000000000","message":"Patch Set 15:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5269/.","accounts_in_message":[],"_revision_number":15},{"id":"a4db2c6f0051238889b10e7fd0d6e023193b301c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 16:10:46.000000000","message":"Patch Set 15:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/0246e4f150234655808c05c0cc76174e/","accounts_in_message":[],"_revision_number":15},{"id":"a382e41e1d035fb95515ae97844a44e9b43fe17d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 16:16:51.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":14},{"id":"1fa46a9acb07846597258b882d33724142ed3f95","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 16:24:45.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":15},{"id":"00414aa9f14d7a86c635b4889f074344eb1fc2de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 16:54:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0246e4f150234655808c05c0cc76174e\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6bf15ab1bcdf4c05b3be3ca4617f632d : SUCCESS in 41m 35s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/02f24828cc2d4383bbca41d8828e31b4 : SUCCESS in 33m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a371aff501fc425f883288919afcb7f1 : SUCCESS in 46m 37s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cd8c077dd9964c72b07c231f31c14662 : SUCCESS in 40m 46s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43a810dc7adf47bfa9676ea371b96cd0 : FAILURE in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f56d95ebc38005c4451cd1ff8bb680a921b001b6","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-26 17:19:44.000000000","message":"Patch Set 15:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1317 : SUCCESS in 0h 50m 8s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":15},{"id":"249754e5fbb1cfcb9ceb977b12180b1f7ed9d557","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-26 17:22:28.000000000","message":"Patch Set 15:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/955 : SUCCESS in 0h 50m 54s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":15},{"id":"021e2c6338a8ef8eb63769f1117eba81b6f81724","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-26 17:22:41.000000000","message":"Patch Set 15:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1048 : SUCCESS in 0h 51m 16s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":15},{"id":"efa0cafd51b40694e573d75724522bc0b120b8e3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-26 17:22:44.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/79e168efb38e46ce9673c6b55d7657e6 : SUCCESS in 48m 20s\n- cinder-storpool-tempest-iscsi-multipath build/6ecdd030d9ef426786e088586bfb5d28 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":15},{"id":"d013d7071c9d377bb290a73e9a309d54d5fa7a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:28:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be7c67fee6a24616a64212621a129420\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc966a36411c4b0998d342ea8db30322 : FAILURE in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a9dedea78aa43b184bac60e0e537082 : SUCCESS in 7m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/251b47bff2344a58934487cd8d93214a : SUCCESS in 10m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29fb160445a9420a80fd5b6d2cbf38eb : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c007f57e1514853afce7ce0aa245875 : SUCCESS in 14m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/79154227f1554bd9bf08137da178bcdb : SUCCESS in 49m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13d35fd547f340dda4b744c818d7db5b : SUCCESS in 1h 02m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8868256418c4be5b9bdfdc40dac4174 : SUCCESS in 1h 32m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e487dec4d9945de9ee4973a56a8c864 : SUCCESS in 58m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f4b299dd5c840fab82eba2f7da5de04 : SUCCESS in 9m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba3fe199559a4b24a8bdef27f59a1039 : TIMED_OUT in 41m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ffea131a3efe4d889b9bf90c4d6e22ce : SUCCESS in 2m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf63a788561c43359d4ebb4dc639f9e8 : SUCCESS in 9m 26s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/5786cd97c7bd4222a2867624b10f1872 : FAILURE in 7m 25s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e32617acc6c94ae490c100baee8408ce : FAILURE in 7m 10s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/57a431f5ad924cc18eb319ee9ffe17ab : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/47b9dc3d64ec46cebfb665096cdcb6e3 : SUCCESS in 4m 46s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/669a00bf95174e29ab6e0bcddafe710a : FAILURE in 7m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6d88549d24cc4988883829b93f13119b : SUCCESS in 10m 50s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/13a7fe1bb05844dcbe932fd382a7594e : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba861179aa1e48ff98d226aa2493031b : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/83261a6403ef437789e5127af467f772 : SUCCESS in 48m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14636519691740698d078d149b6b40ba : FAILURE in 1h 03m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85e2a7fb0b67495a9154ca768f24487b : SUCCESS in 1h 46m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/d5b6b2ec31df447e91e1d8bcaa47e1f0 : FAILURE in 2h 59m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5e5676c321154fd8b4083468ef5a8098 : SUCCESS in 1h 19m 39s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/444544151377450784e7b39ecd43d45c : SUCCESS in 18m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/eee4c1a217fc49699406eab7474b046c : SUCCESS in 1h 17m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/169015fe8b2449a0b105e4cac6ea63ec : SUCCESS in 1h 37m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0939197bd6584f48a1249cfed252baa3 : SUCCESS in 45m 44s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78a5644910c345f3a78e05286e3a37ed : SUCCESS in 1h 29m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/871ce93209b44ab79fcb7d27b21770c6 : SUCCESS in 57m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"8406ce16875ffdb51d9dcd768b4ecac24f6fbd1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:40:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e4cb67990d45afa902580c7ded6e1e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b43072f1802a436fb75c82029a77e7d8 : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/360f2f727460408580912418c59f39ef : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b6cf8bd03eeb427fbd30a21fde40addb : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ae4716b29c38f4178271dd6ae0c6082a3b867166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-26 21:47:51.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"28f9665af228bbf231fc56488ca61cda4bbbcd6b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-26 21:58:45.000000000","message":"Patch Set 16:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5271/.","accounts_in_message":[],"_revision_number":16},{"id":"0d734cd48dbe6eb0c4ddab4e820c256937f974c3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 22:25:59.000000000","message":"Patch Set 16:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d5bdfdda52cc454db887d4eea620aedf/","accounts_in_message":[],"_revision_number":16},{"id":"43343040366d6a67d6c120ec80f1fdf4b64158f4","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-26 22:36:43.000000000","message":"Patch Set 16:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/957 : SUCCESS in 0h 48m 25s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":16},{"id":"7f5472804c21ea027e9a0bdfcaa00fd36111e86e","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-26 22:37:19.000000000","message":"Patch Set 16:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1050 : SUCCESS in 0h 49m 2s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":16},{"id":"c9cf969061f8603eab327c09eb07f6edac62a20e","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-26 22:37:42.000000000","message":"Patch Set 16:\n\nHPE AlletraMP iSCSI CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1319 : SUCCESS in 0h 49m 27s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":16},{"id":"1d3d1e5bd386715f7c074888ab0cbde1d4fb3d55","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-26 22:41:44.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/1709b572e15048169c6b235955324220 : SUCCESS in 45m 57s\n- cinder-storpool-tempest-iscsi-multipath build/e2124f602bd649e1bc42fc387181e2b1 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":16},{"id":"2eed3b89fdf1ebf53e7694cc89ec88e8c78a898c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 22:42:09.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":16},{"id":"5552b33c9c333060d694d2fec96d19a95acc397d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-26 23:11:08.000000000","message":"Patch Set 16:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d5bdfdda52cc454db887d4eea620aedf\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7c5d9260153426e93a552fe455c62ac : SUCCESS in 37m 51s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/104f38c0c813409981e0e65939385506 : SUCCESS in 38m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cf2ca1ba62eb4db69a4231239777faeb : SUCCESS in 46m 56s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/752742d30c974507913ac3f028de83ba : SUCCESS in 37m 38s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f797b5fdd46e4b658745986728d10e7d : FAILURE in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a4a895f16fe260b1f0193b66941163ffcae8a0cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 23:54:31.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a01b0cccd8b5422bb19b9039769bbfe6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348cc3c10f4644d2953d47515f45c2e0 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/284c443c68484965b099cb25917618f2 : SUCCESS in 11m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/443d8faa191142038bca0d2506fc0910 : SUCCESS in 11m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f38a0d2d276d42989f6a05afdad7d71a : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c7e2863afd410fbb9d0212b8dc2192 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b1101fee5ba645a6b3dda707fc87f668 : SUCCESS in 50m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c78ebb139b74281bd8af2515ca56845 : SUCCESS in 1h 10m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88a71d9ba9ef4c00a79999baff02670e : SUCCESS in 1h 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5e91c3f63454fb396f9a922a3f85e09 : SUCCESS in 30m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae122b070824062b7b1b48af81d3dea : SUCCESS in 9m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d11791787efd45e48d5c205e5743c5aa : TIMED_OUT in 41m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c09fa5500696475693f56157bb263cb9 : SUCCESS in 5m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01f1df0d21c2427aabb324eb02675b6c : SUCCESS in 3m 12s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c6b57689d5dd42af8c02319215b18329 : FAILURE in 9m 33s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7ad8ef7d6d00471fad6c5a14cd3654be : FAILURE in 9m 57s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/870cde8d75884475a196af09b3531aa8 : SUCCESS in 5m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/94b84d0bbaef4cf1978c1e88ace58ff1 : SUCCESS in 4m 59s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d7ab3e1cab954167b225528ab1a8b31c : FAILURE in 5m 13s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f895980471194cfe9792e347c20eb1d4 : SUCCESS in 7m 48s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c44be6077b34299a1f4f0e5dc7c9c4a : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e93f1f3b553648e5b092319a1473a377 : SUCCESS in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/01ee5eb995a54058a49296c2fec5a489 : SUCCESS in 48m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/73ad7fbcd16f4275bee60532f1152501 : FAILURE in 1h 03m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c1a8769d739b4309ac8baca33dba894e : SUCCESS in 1h 19m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/aa15649239544825b68d37cf6103cff5 : FAILURE in 2h 04m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/c2985f2264f8429a94e7e75f44a33a01 : SUCCESS in 1h 17m 54s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/c9dc798d08454c91a305e5a6ae33d14f : SUCCESS in 35m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d9ba20ed30c84b2797c16064804b9f5c : SUCCESS in 53m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8f17de3cf8384d47b0be7c98dd8baa9f : SUCCESS in 1h 23m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5abe54c76e0646009792cb64fb3c7a83 : SUCCESS in 44m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/499e76b592d4419db7d3e17baf5de022 : FAILURE in 1h 36m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a09e1c25a74e42858ebfb2a0d4f959 : SUCCESS in 54m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"e84212d03efe6998e05474af6c0da3b5e8cea4ee","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-27 00:11:33.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b179d6275ee89a280ac73b06eb2d338135048e0e","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-27 00:48:10.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-14-fc : SUCCESS in 57m 33s\n- hitachi_ci_nvmetcp https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-14-nvmetcp : SUCCESS in 1h 5m 33s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-14-iscsi : SUCCESS in 1h 18m 8s","accounts_in_message":[],"_revision_number":14},{"id":"510897b2f02b0731f22e73146ceed8e123b62cf0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-27 01:00:20.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5915af4c1f164790b0aa349019d2a5af : SUCCESS in 47m 28s\n- cinder-storpool-tempest-iscsi-multipath build/1e2b7b7bc39f465dac451f9b2eee3d12 : RETRY_LIMIT in 15m 53s","accounts_in_message":[],"_revision_number":16},{"id":"e7fc16f4a9338e3989768a344916023ac36c3ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 01:12:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f845cb221c5b444f8ec657c0ca3aded4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1ee6892385c04aeba4afa6952e93f106 : NODE_FAILURE Node(set) request fa67655eea0949eeb8722a7f56d897d9 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44887fcc03484df88667724285e4e8b5 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0158af0716a44823885bfef1821ecb57 : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"39ec43540711b3290ee4190491875847e52b7373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 02:14:34.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4695ba4eaa485985012ea6bea0af07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b774744a0d473f99d5b71722e07681 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45b1197555e549b8ad58151e03de4ea6 : SUCCESS in 11m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df69d5da76b7402a96680e1df7356036 : SUCCESS in 8m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/47a102f8e946407b972800ef7e9ddc5a : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5ff919c5a54204a0fc9daadf918f05 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/416b1eee948b4b5889b346daa2353786 : SUCCESS in 33m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/62d8344e2e9c42c79b3f48e829c3322c : SUCCESS in 1h 08m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d9ef69b66b0483abbba557152db538e : SUCCESS in 1h 31m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c19db03803b046e0a95a51c283daa006 : SUCCESS in 45m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41bd1a1db046417ca23ba652589bc5fb : SUCCESS in 8m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f87c3f4205324658806c78875d2cb9cb : SUCCESS in 18m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a18b1b9761e4dea85f05cd63fff2dfe : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23e452db32ad48118328febbba3f67bb : SUCCESS in 6m 06s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/6bab48aa065c4a2f9743db7d36267055 : SUCCESS in 11m 54s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/05e0c3a6551345f4beedb7b488699789 : FAILURE in 6m 48s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1177b7ddd7c476e88ecf064391cd0d7 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/010be365282a4b2d80673e926434cf95 : SUCCESS in 5m 59s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bbbd945ebfec45089939c4a7df2877fb : FAILURE in 3m 41s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/db5ac672d72f44f6aa4882a91e737409 : SUCCESS in 6m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9edc848c8a1a4b5bb67fd86c8af26ffd : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13595505e7c643e9a5ef76944ee7915e : SUCCESS in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ef1318ab37743f7999971f7d9566615 : SUCCESS in 1h 45m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48f970d07d6b4aea826802e5970868e2 : FAILURE in 1h 27m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/18e09aedfe454906a2c5e72fb99233fe : SUCCESS in 1h 44m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b2d4b8629e8445ec838edb2faf90d220 : FAILURE in 1h 58m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/3c599368388b4876b7c5f3aad73b2654 : SUCCESS in 1h 16m 21s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f1e8c5ef9f4c447e8aeb0bee8671804d : SUCCESS in 38m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0829c5ea41f04252bc6e4a9fb07a09fa : SUCCESS in 38m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5dfceb3f358a47aa8941f56577e1e41b : SUCCESS in 1h 23m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0493e2ea71bb4c7f8180399755280411 : SUCCESS in 47m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a1ce1574060f42308509e1e3857f3baf : SUCCESS in 46m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f63aae7057f0439ebc3de20870713d5c : SUCCESS in 26m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"3d119b2d0cccf6ae072b7c7b7b4ecbba010acc58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 02:28:35.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_08_2026_02_21_09\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"b4d4d8c34816f9ab6f10eaa6ca7e13038f141ab1","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-27 03:26:17.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-fc : FAILURE in 9m 43s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-iscsi : SUCCESS in 1h 3m 37s","accounts_in_message":[],"_revision_number":16},{"id":"ca21a055928d36bb7c0352b0e60cc2c71316735f","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-27 05:58:32.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-fc : FAILURE in 9m 5s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-iscsi : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":16},{"id":"d7aca50001a26868145954212027591084b202b3","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-28 06:10:00.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-fc : SUCCESS in 1h 6m 49s\n- hitachi_ci_nvmetcp https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-nvmetcp : SUCCESS in 1h 15m 43s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-16-iscsi : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":16},{"id":"0e87f5e27e63f31be602f3eaea8f380c51a8ae36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:10:06.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"9ec10d42bcf2631431251c31e339f89e2afcb9e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:10:12.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 997469,17","accounts_in_message":[],"_revision_number":17},{"id":"05528e4f87d4310bab26aca10fcf942d847c774e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 22:10:17.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev/openstack/cinder for 997469,17","accounts_in_message":[],"_revision_number":17},{"id":"a461b6a7ba5da904702abc1f596ab6da7806b5c5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 22:10:20.000000000","message":"Patch Set 17:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/cinder for 997469,17","accounts_in_message":[],"_revision_number":17},{"id":"37059d67bad7b025246047b62e3d138ba8d143a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:10:49.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nThe following files contain Git conflicts:\n* cinder/tests/unit/volume/drivers/hitachi/test_hitachi_hbsd_utils.py\n* cinder/volume/drivers/hitachi/hbsd_common.py\n* cinder/volume/drivers/hitachi/hbsd_rest.py\n* cinder/volume/drivers/hitachi/hbsd_rest_api.py\n* cinder/volume/drivers/hitachi/hbsd_utils.py\n","accounts_in_message":[],"_revision_number":18},{"id":"3eae03ff55b471f0bc256bb7bb18c57079232bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:10:50.000000000","message":"Patch Set 17: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 997469,17","accounts_in_message":[],"_revision_number":17},{"id":"cf175080ffdd1b61df1c3f07a785e1592679c0f7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:11:41.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7a643380ea186d54a32bdfb30d64ba2a5d798c9b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:18:40.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"f372f37fcbcd8f46f68f735001588bb4436bec06","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:18:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":19},{"id":"e653a5df01a8f03ce88fa38e488272d0676f861b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:31:00.000000000","message":"Patch Set 17:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5309/.","accounts_in_message":[],"_revision_number":17},{"id":"3c3d595dcab85e96c27255cb58dfcdc261594379","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:34:24.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaac671c115f4821bcf501df2cd140d1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e983004a0dd4cdfb0398976f7c1b85d : FAILURE in 9m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/852f59133ac54a70a4cf894a2479ceda : FAILURE in 9m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f25e602d4da9499dba1cede34a946207 : FAILURE in 9m 20s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"d1db0a600289700252b3c80833f3b21cbac701ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-08-28 22:36:45.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"97300b19637adddd1c32e37f42b506057d84ef08","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:40:36.000000000","message":"Patch Set 18:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5310/.","accounts_in_message":[],"_revision_number":18},{"id":"39323713f4f0d39ce5bac2738f3f8366e6a9c062","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:49:05.000000000","message":"Patch Set 19:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5311/.","accounts_in_message":[],"_revision_number":19},{"id":"6ca80d0a9047524212c43c965479f9d0ee91a506","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-28 22:57:53.000000000","message":"Patch Set 20:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5312/.","accounts_in_message":[],"_revision_number":20},{"id":"ffe18c3f51fca18c8b6ccf205c9a232d7ebc4bf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 22:59:22.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f59880d916d482d87f0e9e5f244bae8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/900479330e6e4dbb81ddfe330eb33894 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0cba6efc314494bb7f90205bc42fe3b : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7ddcdf57ac1340818deecaa8ac5178fe : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f18df822df828dc93ca850e4314048929d6cb863","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:04:40.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":18},{"id":"d9380a91af958038cd7384fdf6a3509482788957","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:09:56.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":17},{"id":"3055d13b655e9ccb567bfb3fe80fc6a56f9d319f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-28 23:13:27.000000000","message":"Patch Set 20:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/192b6e964c5f4259aa4889547ccde85a/","accounts_in_message":[],"_revision_number":20},{"id":"93252d612b6fb496dc6a92595910c9830ea54fca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:17:33.000000000","message":"Patch Set 19:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":19},{"id":"64df04654328da1bf600b64acec5af3465c8613d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:30:58.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":20},{"id":"84411bbe67417504b672a199d679cc6c5655d015","author":{"_account_id":39082,"name":"Vijay Jatti","display_name":"Vijay Jatti","email":"vijay.jatti@hpe.com","username":"vijay-jatti"},"date":"2026-08-28 23:42:12.000000000","message":"Patch Set 20:\n\nHPE AlletraMP iSCSI CI\n\nBuild failed.\n\n- OpenstackCI/Openstack_iSCSi_CI build/1357 : FAILURE in 0h 37m 8s\n\niSCSI CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/new_logs/997469","accounts_in_message":[],"_revision_number":20},{"id":"35c376f6d137db669118f6f97b1a63ace0a6b64c","author":{"_account_id":39391,"name":"Rashi Jain","display_name":"HPE AlletraMP NVMeTCP CI","email":"rashi.jain@hpe.com","username":"rashij"},"date":"2026-08-28 23:49:33.000000000","message":"Patch Set 20:\n\nHPE AlletraMP NVMeTCP CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_NVME_CI build/996 : SUCCESS in 0h 45m 37s\n\nNVMe/TCP CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/NVME_new_logs/997469","accounts_in_message":[],"_revision_number":20},{"id":"0b48895cb83a7db4ae372bcf345e49e83d1d4bfa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-28 23:50:56.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/fd2991e5936241fbbeeee717cd2d7417 : SUCCESS in 41m 42s\n- cinder-storpool-tempest-iscsi-multipath build/8bbb73c970ad45eb9e9b0e08142be9c6 : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":20},{"id":"2e4b65db0d4990ab407f85b06579534b18b0d32f","author":{"_account_id":39323,"name":"santosh-kumar vidiyala","display_name":"HPE AlletraMP FC CI","email":"santosh-kumar.vidiyala@hpe.com","username":"santosh-kumar"},"date":"2026-08-28 23:51:57.000000000","message":"Patch Set 20:\n\nHPE AlletraMP FC CI\n\nBuild succeeded.\n\n- OpenstackCI/Openstack_FC_CI build/1088 : SUCCESS in 0h 47m 30s\n\nFC CI log: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/FC_new_logs/997469","accounts_in_message":[],"_revision_number":20},{"id":"9a3dc202a3fb31f14a0410879213bb7c4d4d3614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 00:47:05.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb409877e8a445d6903459ae4728d4c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3d9cb8779942238f17a574f886082b : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad0a2245062048b9b01a41b64089e06d : SUCCESS in 12m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b202973cba3406b9419605c12847781 : SUCCESS in 10m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2473f26ece144f3bcf368f3005b0683 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0311505bf3f440fa800d622d0c9e947 : SUCCESS in 6m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d53637321fe847708b92a0ca8b79164b : SUCCESS in 1h 10m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a05e869f3e474edcacbaf3d10388295c : SUCCESS in 29m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9624fcb51eae4879b754761d66a65a3b : SUCCESS in 54m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/528d68606e6545ed810046a61bcbba0a : SUCCESS in 57m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5df522ed98d453ca6f94ed11af9ee73 : SUCCESS in 9m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/63dd33951279443a96730d8da1978602 : SUCCESS in 21m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/304f2ce3d6ff4007b5c1fef3738671bf : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd559d62a6d94e07904e361a289b2f78 : SUCCESS in 6m 40s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2f04b2d1abaf4d31b96ca9acd0c4adc6 : SUCCESS in 8m 32s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ee1ac613517744d088f6d048e345619a : FAILURE in 7m 28s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/766acdfac7864bcd84eac128ca925038 : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dd16e65aaacb4f878b423560376e5e97 : SUCCESS in 8m 48s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a8be5dd532fc460388961cf2186ae2da : FAILURE in 8m 09s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2dd7c62cf3574a7c9f56030ba2ee8230 : SUCCESS in 7m 21s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6b96b6027ab84d718cd2ede06a215462 : SUCCESS in 34m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4109801fdc594761a20bbb163fee135e : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e4598ced031456897710d0231b1459c : SUCCESS in 1h 28m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3845f9b17d2b45529ce9482aeb7832be : FAILURE in 2h 04m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eff498da5b3a475aaa42b6b96f753aa8 : SUCCESS in 1h 28m 38s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/09622bdbabba4f589e69a83d17979091 : FAILURE in 1h 37m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/9e6c746051474b6f9c767b6454ff3ec2 : SUCCESS in 1h 44m 43s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/53a4a0aebb0341cd97d9aa9e64e8826e : SUCCESS in 37m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a791373950da43ddaaef7fd04c21754a : SUCCESS in 36m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/90c5a022817b4ebfa93bece00f3bd43b : SUCCESS in 1h 26m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/39f554e501964b4e9b41e1adf91e35a4 : SUCCESS in 44m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1bd8320b22b47a4a9c19c475d35d49a : SUCCESS in 1h 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/632141dc8e1c4479a173bfc5a6e58357 : SUCCESS in 58m 11s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"e94eb1860fd017e6d3bfc6d60a7c2edd56651f2c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-29 00:51:17.000000000","message":"Patch Set 20:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/192b6e964c5f4259aa4889547ccde85a\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e9d7fe5f5de41428b79d0378f8c21a0 : SUCCESS in 44m 43s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7786ebaa596840e2b9d1eaed52922717 : FAILURE in 1h 41m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a7386f4107b14cf19096eebf1490f16f : SUCCESS in 45m 45s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ba5149c4343b44bca311ea558ad3b067 : SUCCESS in 32m 54s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cbc7b1e1685046819e174a75ed514582 : FAILURE in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a996ff7bd98fd2d6c790c9a5d15add5b8009b7d0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-29 00:57:10.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_08_2026_00_50_37\nLeave a comment with","accounts_in_message":[],"_revision_number":20},{"id":"281335e52f80f6c29a4697817b8395b68690cef2","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 02:38:04.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-20-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-20-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":20},{"id":"de4186ff6f5a50c468f7e08f83d98ee7599b77af","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-08-29 19:31:56.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-20-fc : SUCCESS in 1h 4m 59s\n- hitachi_ci_nvmetcp https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-20-nvmetcp : SUCCESS in 1h 9m 46s\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-20-iscsi : SUCCESS in 1h 13m 1s","accounts_in_message":[],"_revision_number":20},{"id":"da79b2b53850981d6d940ec462cb512c68488ca5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-30 05:14:52.000000000","message":"Patch Set 14:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5092/.","accounts_in_message":[],"_revision_number":14},{"id":"72c0ada23a8f7726571af9bdce32e6a47bb14f6c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-30 07:14:11.000000000","message":"Patch Set 15:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5093/.","accounts_in_message":[],"_revision_number":15},{"id":"953ae406e9c00099785a59fce06f904f9b5211da","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-30 11:39:01.000000000","message":"Patch Set 16:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5095/.","accounts_in_message":[],"_revision_number":16},{"id":"a28e939970fb5da5a17437f17f220dd7c08c5900","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 03:16:46.000000000","message":"Patch Set 17:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5133/.","accounts_in_message":[],"_revision_number":17},{"id":"52b266d70d76bf3e837179f39078a097f17a9766","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 03:54:32.000000000","message":"Patch Set 18:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5134/.","accounts_in_message":[],"_revision_number":18},{"id":"9861b3cbb6b9b52d555c3fa49a80bd7e39031d91","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 04:32:17.000000000","message":"Patch Set 19:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5135/.","accounts_in_message":[],"_revision_number":19},{"id":"81f87075d35efe5009a5460964d421f39d61a90f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-02 05:09:07.000000000","message":"Patch Set 20:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5136/.","accounts_in_message":[],"_revision_number":20},{"id":"d65576ca6a67f2aed6b4b6106648cd744ea39433","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-10 23:56:06.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-20-fc : SUCCESS in 1h 4m 59s\n- hitachi_ci_nvmetcp https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-20-nvmetcp : SUCCESS in 1h 9m 46s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-20-iscsi : SUCCESS in 1h 13m 1s","accounts_in_message":[],"_revision_number":20},{"id":"d4ff51bb9b87aee4c1e44d4a74fc59d6dcefb6aa","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-11 09:59:13.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c22b540f71ac33ccf8f0133d1ed886de125353b9","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-11 10:00:16.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"67cf48fddc50c21e5c0675f6113626c5ba59093b","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-17 03:41:50.000000000","message":"Patch Set 20: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"2cca222762a3f80db45da7cfbb2ca898970a7dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-17 23:28:13.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"ddf6b539d1ff597e4566a3bc83619b752b72a1e3","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-17 23:28:58.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"610092d648cad8c21debbb9481e5a05bba964793","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-17 23:53:23.000000000","message":"Patch Set 21:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/18b9b417e69e47fea4bbda93ab7410da/","accounts_in_message":[],"_revision_number":21},{"id":"081ced215d1da873730a494ec1237b44112d2ed2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 23:53:30.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee3aaf24f3d4aba858c7450524af49f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5aa8942cbec64da29f181fde4fb238f6 : FAILURE in 15m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec6d4a31c63642cb9e3b317ec4cfdc13 : FAILURE in 16m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9f50f51dd88e457da51675c9f26db631 : FAILURE in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"71b556617fbf13539cf54b6d7bc8c05884dbd644","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 00:19:24.000000000","message":"Patch Set 21:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/18b9b417e69e47fea4bbda93ab7410da\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6293441b1cfb4b7f9333a96900c35dd4 : SUCCESS in 46m 06s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/29612c9113fa4766ac8cc0d9e7d5a990 : SUCCESS in 32m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d339f7bc0e7f4c4982f82d8de5781102 : SUCCESS in 50m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8e7153aab3d54631ae850fc24c6027d7 : SUCCESS in 33m 06s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/fc264ab0141042ec9cf72da839880d70 : FAILURE in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"eb9f47fb7a34832296b18f22a8b3dee6ff9974d3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 00:20:57.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":21},{"id":"927fe827488a0911feabbe0a5654bc978330e6e5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 01:34:49.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/52e8868b56b64d31afd92ba906a4628c : SUCCESS in 48m 45s\n- cinder-storpool-tempest-iscsi-multipath build/5d259d1f68f740f5a676594ae82ddb57 : SUCCESS in 49m 51s","accounts_in_message":[],"_revision_number":21},{"id":"6de7548b47de13562fee205ecce00f9818e91bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 01:52:20.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dffdf865d77d41528a2d81dbec78ea83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd372db7fc7f4e0cba799b98a0b29459 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dd8621de8e84675ac660cc9269a30ee : FAILURE in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e945ca056d064977a66bf390a8d9ea92 : FAILURE in 7m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3852d6d9b44047f6a8f57e8019a9f144 : FAILURE in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc13aba0c614a929c97896e9ac87f82 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a5f9ec0754eb415e90184dbc2e8ec9bd : SUCCESS in 1h 06m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3437869dae3f4a7eb282c3b996a74a70 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b1e67dcd38b4a30a5327b0557bba8b9 : SUCCESS in 1h 24m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed7c505e8ab7436f9965eb7f07b4a795 : SUCCESS in 31m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/065d867327b54839af11a055639e1d82 : SUCCESS in 4m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29ff01b109a64e4b8e8e6a66baf72f66 : FAILURE in 17m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4bb6a65eb1a14f8d97e0ea2bcda95ba3 : SUCCESS in 3m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9e87bc1df6141bfb44d942bf847d612 : SUCCESS in 4m 04s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e48c78bb553d46f1b9375533895fb4cc : FAILURE in 7m 57s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9acfe849859347429421c03e85936db0 : FAILURE in 8m 03s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/50a60c9088264c13bcdf26a23eeaf1a3 : SUCCESS in 6m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8cf0cdab787e4d85a9949fca9d1f2c8e : SUCCESS in 5m 30s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6ab15a0e6919460f9761d2285ea03855 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/9ea22186513c4257abd83de491eeb9bd : SUCCESS in 8m 29s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4aa96f060e9e4901a724d8cf8f8497b0 : SUCCESS in 38m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7ee3a655f1041e1800c1e5fd6b80cce : FAILURE in 3m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c1188155288d4b3cbeae81d05dde55ac : SUCCESS in 50m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ae16bf5a72a148a492041b0c5bf91ae9 : FAILURE in 2h 00m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ce07b33f45f040d4b71ad9b54dfb2906 : SUCCESS in 1h 50m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/85757ca71a404fedae7e3915200a4f51 : FAILURE in 2h 21m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/6db4eb572a6d45e5966c3ca26ecb4180 : SUCCESS in 1h 55m 40s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e78335489d744d1eacdbaae4b84099c9 : SUCCESS in 24m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1bf47f3a18da4c6e9b984e09e4061b67 : SUCCESS in 1h 17m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2e65265714714f9b99699ab191e03d31 : SUCCESS in 1h 32m 39s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e62476b3497b4e4a9a61c2f108f1eedc : SUCCESS in 27m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15bc38f0616a4cc1a72ab8d71b209708 : SUCCESS in 51m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f454c3a8a7ba43b6a9287a4d0417a9b0 : SUCCESS in 41m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"7223507d5fe67744d7253bf25b664e09e2eb7311","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 02:05:50.000000000","message":"Patch Set 21:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5592/.","accounts_in_message":[],"_revision_number":21},{"id":"a0f7acb98149315e6ac81e7b4edb2c4c0ee5d6c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 02:45:18.000000000","message":"Patch Set 21:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5415/.","accounts_in_message":[],"_revision_number":21},{"id":"e2d791f51da9fe726b51ac573bc1755cfb98e8ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"date":"2026-09-18 03:11:17.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"73d0cf17306d6a845af2612a89c30135b22372bf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 03:30:24.000000000","message":"Patch Set 22:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/b13df71268234782baee38f0a2931819/","accounts_in_message":[],"_revision_number":22},{"id":"fda75952fc8993134243b1de0d37831a5cd001f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 03:50:37.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27602a68af54ff7a0699c30414581a1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa7641f4306245bd9857fd97042b7325 : SUCCESS in 17m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f70ed38869442d8ac7417464e3adca1 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a8766e055d64182b1cfa760378519f4 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"94173f8b7bd80abcc78c8dbcd8022f2599069d2c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-18 04:00:20.000000000","message":"Patch Set 22:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/b13df71268234782baee38f0a2931819\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c02fe8c434ba49a6a1c31706ff3e43b2 : SUCCESS in 40m 34s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5665ef998bc44a3294ba28c1f191a740 : SUCCESS in 36m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/344aebc2c4244041a4d5bc30cc6aa453 : SUCCESS in 48m 31s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d679a117436458894e13642ade0cac9 : SUCCESS in 35m 10s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e8a05477e87946888b9777690e83ebc6 : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"940a0882ff07d66d76857c72364bb088527fe87b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-18 04:09:25.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/fb25c4cc716e423d98e1679e4210da81 : SUCCESS in 46m 08s\n- cinder-storpool-tempest-iscsi-multipath build/01dd494543624a42bf6afa5e7ec9fe1a : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":22},{"id":"7f4dbe7ecd56bec0ae405ca20a7b99934c1b9465","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-18 04:12:21.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* quobyteci:  cinder_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":22},{"id":"55095487c4c349875a77c0cdf275bf78d544fa17","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 04:17:56.000000000","message":"Patch Set 22:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5416/.","accounts_in_message":[],"_revision_number":22},{"id":"2ca57dd079b990e0115be51b51385ca481cbc01d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-18 04:18:12.000000000","message":"Patch Set 22:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5593/.","accounts_in_message":[],"_revision_number":22},{"id":"9701d4c110faa414b8a7431536d31fa474b8788e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 05:36:04.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd1c2a8a2744499aaabc7f561f8744a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296b382daa97479da1f8cb0d5e1c4e45 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2b263b8317240b5ad16bd12b0208d6f : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ecd58619fda4456b4fc9cd8bdcdfb62 : SUCCESS in 11m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2723fd8bdf943f99c0447bf65519efe : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314f05843b26438da06cbfe8a350b9d4 : SUCCESS in 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/685d34700d254df98abd7666fbfc53ea : SUCCESS in 40m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d49bd6e3620e48e9983ce828737f9575 : SUCCESS in 51m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0327198a09ff46eaa9f3e3e503943189 : SUCCESS in 1h 35m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37d0fe939faa4cddae02308c07630c81 : SUCCESS in 29m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0a00aa1893045d29b0910b6bcfbeb04 : SUCCESS in 7m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/31ca15b7cec649d4ab787d1b640b2d5b : TIMED_OUT in 41m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/58839719c0b34176ad30c51bf754b104 : SUCCESS in 3m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb7c577cc53645289334e08eb42ccd84 : SUCCESS in 3m 33s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/13592adaba894e2ebbf6c07819f7adea : SUCCESS in 13m 58s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3f0f0357a270430595576842f641de49 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/82100d160428451580441d853cd5630f : SUCCESS in 7m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a95962357a9f42de94b7b1390de6a2b3 : SUCCESS in 8m 22s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/96b34b6c61354104a16c6a526f9af4d0 : FAILURE in 5m 21s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/65d2e7e86e714f4bb50c5d5540d16dee : SUCCESS in 9m 40s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/025e4f7c4ae44ead817c04d71c82cd2e : SUCCESS in 22m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/268be20aafb041faa74f3295ae990f6d : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a5e8674133546d8bd327c9bdfedd388 : SUCCESS in 47m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/878418bd9b644e3cb1c8407da89d654d : FAILURE in 1h 59m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/840345141837478fbf25a525b0735b8c : SUCCESS in 2h 09m 31s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/678f9dc5c6e048a19880b0439c700ed6 : FAILURE in 2h 22m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/04df122589a845189305382e576bfc50 : SUCCESS in 1h 24m 36s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/174ca0a7138947148b1daa8ca1b19755 : SUCCESS in 29m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9782a33ab26f418e8d4b090449192ed3 : SUCCESS in 1h 14m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/656352f3fa5c4fb8b6558d31d9fb8e37 : SUCCESS in 1h 40m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5c5ec531278542fcb31d3df6afd02d0d : SUCCESS in 43m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b6d9fd384da4cb19ee7548b50acc412 : SUCCESS in 1h 11m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/494c38e8e7304e8f88d815b4dc9d0616 : SUCCESS in 1h 01m 01s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"201955274f8e6caf0489a11847ca53b6121d6107","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-18 05:45:44.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_09_2026_05_39_41\nLeave a comment with","accounts_in_message":[],"_revision_number":22},{"id":"6a82ec84983d3158057f63602debf3269936ceb8","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-18 05:46:56.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- hitachi_ci_fc https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-22-fc : NODE_FAILURE\n- hitachi_ci_iscsi https://github.com/gamboa-hitachi/Hitachi-CI-Results/blob/main/refs-changes-69-997469-22-iscsi : NODE_FAILURE","accounts_in_message":[],"_revision_number":22},{"id":"0de58b5503f24bc02c3325b1edf042b55c26f447","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-18 13:11:58.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-22-fc : SUCCESS in 51m 6s\n- hitachi_ci_nvmetcp https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-22-nvmetcp : SUCCESS in 1h 45s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-22-iscsi : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":22},{"id":"c4a8e05f8073adfbea1cf562c09b0d62fdb24530","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-09-18 14:15:10.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"cc916b0ca53c6ef44473e98a17a49996e01041af","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-24 05:44:49.000000000","message":"Patch Set 22: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":22},{"id":"8a03f843a4d0bb282a2f6c2840c0898453dfbc24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 09:36:54.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"4c9fa22a75000ad7963ec8c36c89e9a85d6ac345","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 09:37:04.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 997469,23","accounts_in_message":[],"_revision_number":23},{"id":"c9a50b1c2d0bbc19057bc4f9e19f15a69e692da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 09:37:04.000000000","message":"Patch Set 23: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 997469,23","accounts_in_message":[],"_revision_number":23},{"id":"e821900601806a1c7f320511e629a9019ec785b2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-26 09:37:06.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev/openstack/cinder for 997469,23","accounts_in_message":[],"_revision_number":23},{"id":"7e2fa92d06893d09ffe56cc31eca3c21b3f6f971","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:08.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"98634fbd724094115aa7cbd5f5ef829c7d7edd23","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-26 09:37:08.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev-third-party-ci/openstack/cinder for 997469,23","accounts_in_message":[],"_revision_number":23},{"id":"174dd634de8bf070296acfa7f60728d4a89a9bf4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:09.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"25d43b95dccb2a53412445b855a00b6d48bf0c8e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:11.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":23},{"id":"510301f0942a76b654c51f2980acee91055f6ae4","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:13.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":23},{"id":"0d5821f8d28b43ae550b3106d9b6c9b7fc2c9a16","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:15.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerScale_CI","accounts_in_message":[],"_revision_number":23},{"id":"ade9343c1790070cdc977abe3102bb747d6c083e","author":{"_account_id":37792,"name":"Dell Openstack CI","display_name":"Dell Openstack CI","email":"dellopenstackci@dell.com","username":"dellopenstackci"},"date":"2026-09-26 09:37:16.000000000","message":"Patch Set 23:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging openstack.gerrit/openstack/cinder for 997469,23\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"875e78f28ef2d6177c362e173ce9310a5fd2e4c7","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 09:37:28.000000000","message":"Patch Set 23:\n\n(18 comments)","accounts_in_message":[],"_revision_number":23},{"id":"aa486ba5d8c823967ea897416588481632b05f64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 09:37:52.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nThe following files contain Git conflicts:\n* cinder/volume/drivers/hitachi/hbsd_utils.py\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"8c61f7673b84caa44f6e6cb198a4abf3e4f5f4d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 09:38:26.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"dd6cf85c325ad7a74aff4a6b96a07fc14167af15","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 09:38:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":25},{"id":"6467f6187c9297660b8246dc869993aed68ca27f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-26 09:54:15.000000000","message":"Patch Set 25:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/5469f84f38c0412d9123d8b7fc70692d/","accounts_in_message":[],"_revision_number":25},{"id":"3ac3c6b4f36a5effe03a032ac7b25c6865b303b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 10:00:43.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8860d30b6b364fa68fea7ade9ca82a04\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/42615eccbbeb420799c45c3ec2eb4dd7 : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94ce2fe81a3d41cf9646d8bc459104d4 : SUCCESS in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"26a5156f83d8ace3186baaff25b31502b0f3c3ee","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-26 10:21:13.000000000","message":"Patch Set 25:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/5469f84f38c0412d9123d8b7fc70692d\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/25fa644d8c0a4760a9fe28861278f06f : SUCCESS in 40m 40s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91dc5e31333849caac4f107e14a7f89b : SUCCESS in 36m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/266ace14ff10454983879eb210ccea1b : SUCCESS in 42m 12s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/9941bcd1540b4429b53eef9f1b032fcf : SUCCESS in 28m 27s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/000f61ae3bd94045a1906d062430ef88 : FAILURE in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"1676e0f412c639e3343df5d6a2dc79767f9deb86","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-26 10:29:51.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":23},{"id":"f48096086014770723126635d79f4fc24d714a5e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-26 10:32:28.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/e799d9b56e4f4a9f8f5fad415e095c36 : SUCCESS in 47m 59s\n- cinder-storpool-tempest-iscsi-multipath build/35b11bf8c7424955be58c26f16b7ca1c : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":25},{"id":"8e4989fa2f712806fb84468cd1409250b5eccb66","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-26 10:33:04.000000000","message":"Patch Set 25:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : FAILURE in 0h 47m 6s\n- FC       : FAILURE in 0h 48m 49s\n- NVMe/TCP : FAILURE in 0h 47m 17s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/997469","accounts_in_message":[],"_revision_number":25},{"id":"0fdc888dbeff675f55827a583fc2a3ebfc5a0498","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-26 10:33:32.000000000","message":"Patch Set 24:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":24},{"id":"1c061d5bbf9d5f9ba7fc0213ade4623097bd30bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-26 10:33:52.000000000","message":"Patch Set 25:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":25},{"id":"2a6ba0982a774dc8108d2e8f46d06e21e6c97404","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 10:41:51.000000000","message":"Patch Set 23:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5713/.","accounts_in_message":[],"_revision_number":23},{"id":"2e70950b12f8a39da634f055529085a28e54bfaf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 10:48:29.000000000","message":"Patch Set 23:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5536/.","accounts_in_message":[],"_revision_number":23},{"id":"eb8f1fb112e531c57fe8a0aa6777304ca9e63246","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 11:46:16.000000000","message":"Patch Set 24:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5714/.","accounts_in_message":[],"_revision_number":24},{"id":"332aad825366faa44daceb8023d7589730f049e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 11:56:29.000000000","message":"Patch Set 25: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1d0e76b57094cefb72b0ece2ca4695a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a70ce4a9024cffba9ff2a6818750da : FAILURE in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaadc3b7954d4b5a966b665647bc4378 : SUCCESS in 7m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f27575a5aee04b9293a5a9f832eea59e : TIMED_OUT in 41m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe975e1d6aab46829e2efbedc6430ec5 : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/4128601185f744a78d9b934552fb08ef : SUCCESS in 49m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b98be39c1ad449928ce687a346a34137 : SUCCESS in 48m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8305bc2fdba842acbc9cb664fb4d366d : SUCCESS in 1h 15m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1518c58a4df549ae8634452251f0400a : SUCCESS in 28m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fc978e58dbc4702a9707f0a9cf5d11e : SUCCESS in 8m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/30c589ac03e948cd9894dedaaafd37c1 : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a7b3ad90bad4636ae7bad53806c4aa4 : SUCCESS in 2m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf3e5b9ee80a4dbeba786f332fba6581 : SUCCESS in 3m 48s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c74768229fcc462897a1643a4ae6a621 : SUCCESS in 12m 44s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2cea98a21d7b4a0c8bb63a9ef505d25d : FAILURE in 7m 17s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f85c6af9d068479883dd8c342517b5d7 : SUCCESS in 5m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a8a5ebe8b715426595b7cf8004eeee62 : SUCCESS in 6m 15s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2804cb074a034fa19da79beeabf186a2 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/06b9ac1050534b60aaf0e6b393564b4a : SUCCESS in 9m 13s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f52a9a8df80f457790c1b42ab91cb268 : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/baf6acac7c14499caf031af4d1f598b2 : FAILURE in 5m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a0163602e68422fbda700302a2ea48a : SUCCESS in 1h 28m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f1ffc2c494b849f797cd8c5990544449 : FAILURE in 1h 27m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ee6f4f6d1504b3cbb841beec004a5f7 : SUCCESS in 1h 29m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/cb5e38672ecd499da455afa3539a564b : FAILURE in 2h 09m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/5d12da2bb61d4db19a3a5f2170b00a4c : SUCCESS in 1h 24m 27s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/dc77e15619c040258c7120efd249019a : SUCCESS in 33m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fce8fe09136c460ca8033788ba0553d6 : SUCCESS in 1h 12m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e90e4a07104466fac603f1cfac008bf : SUCCESS in 1h 25m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7e9faf5270446ad8c16ce94409f5f1c : SUCCESS in 27m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a17800067d1343d4a39c33d89d6a23aa : SUCCESS in 43m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0648fbb1dacc4cb5a3ac995e12ed7c08 : SUCCESS in 29m 03s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":25},{"id":"63218609cd9c1587fb8b64f8328a5b169daf2b24","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 12:02:09.000000000","message":"Patch Set 24:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5537/.","accounts_in_message":[],"_revision_number":24},{"id":"79d81f5b3fb90967cc2581b1c1c04d7eca1b467a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 12:53:57.000000000","message":"Patch Set 25:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5715/.","accounts_in_message":[],"_revision_number":25},{"id":"a4f4ef8626152f15ff64bacfe677f6cea79a8454","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 13:04:04.000000000","message":"Patch Set 25:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5538/.","accounts_in_message":[],"_revision_number":25},{"id":"0fbc3b611dc63449b08c714136d8ebc8b90edad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-26 13:12:08.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"6d96ec2327ea3c3e92c158ceedfcdfa480f84977","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-26 13:28:07.000000000","message":"Patch Set 26:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/d748763d49f848a9a80f4f1cde43310c/","accounts_in_message":[],"_revision_number":26},{"id":"a3c1dd84b8d7ca3787182be3f3bda2da31ff398e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-26 13:35:31.000000000","message":"Patch Set 26: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":26},{"id":"3a552c4868cd13e0e27e769a0206366741bf4043","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-26 13:56:40.000000000","message":"Patch Set 26:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d748763d49f848a9a80f4f1cde43310c\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6d046e25b32483fb982db8ebd694d7f : SUCCESS in 40m 46s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/368135effd17426e9b47566973e66178 : SUCCESS in 31m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f5a515e745fa40ecadde43267057a833 : SUCCESS in 44m 01s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d44ce00798f44dba9f0b7aa2a349ea9f : SUCCESS in 35m 08s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/20126f50552b40ab89c395a547fca73c : FAILURE in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c8edc1e12944a9ff2c5170c0c53f1cd451f99e21","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-26 14:06:00.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/ce8d363714d44429b29ab8ceb203d708 : SUCCESS in 47m 29s\n- cinder-storpool-tempest-iscsi-multipath build/4cc8da3dcbf14af5aafba42d0aec4149 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":26},{"id":"734c36e3a912f2abf976d63229645ba1282a3c0b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-26 14:07:03.000000000","message":"Patch Set 26:\n\nHPE AlletraMP CI\n\nBuild failed.\n\n- iSCSI    : SUCCESS in 0h 49m 11s\n- FC       : FAILURE in 0h 49m 58s\n- NVMe/TCP : FAILURE in 0h 48m 13s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/997469","accounts_in_message":[],"_revision_number":26},{"id":"bbd196753499a94602b06b7fd3ccd81194efafa3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-26 14:07:38.000000000","message":"Patch Set 26:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":26},{"id":"bdd6f0c558ec901015d793eed28645fa2088ff1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 14:15:24.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e5b01e61fd4262b7667c609699a861\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab06d5b93952454d909797954bf11b3b : SUCCESS in 17m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d49db40911bd4774ae148c25b3af2746 : TIMED_OUT in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"bbd9c9fa795f89ef47f8b0728391527c30dfeb41","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 14:22:10.000000000","message":"Patch Set 26:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5716/.","accounts_in_message":[],"_revision_number":26},{"id":"ea0171fb959777999f4afbe2e4d03710f077017f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-26 14:23:09.000000000","message":"Patch Set 26:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5539/.","accounts_in_message":[],"_revision_number":26},{"id":"ee9455552d7c1633260aedc22069dd7fb58dff43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 15:45:14.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe011ae33394c528eead71213563239\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9542c7865c20475fb3b0b2e2baa1d47d : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/247f5d2ac5aa489c9ea885107a6eb780 : SUCCESS in 7m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aaa155c5507245fcae2b05bb23e2d8f2 : SUCCESS in 14m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306c19abf6ce418a85239517edf82dc0 : SUCCESS in 14m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b9c5da3bc48842f08576b6fa92df5649 : SUCCESS in 30m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0443850e7acf4f0d986006a3c69aca23 : SUCCESS in 46m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/162ca29cea734455b72fa06564471257 : SUCCESS in 1h 21m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24a0fc811ddd45cb8b5cdaec367f6939 : SUCCESS in 1h 06m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a380f6add88445da96e6544d8a2433da : SUCCESS in 6m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54a0a7b1ff9243b9893d1293dea15e1f : TIMED_OUT in 41m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6b86608729674f278fb0b7893c176f97 : SUCCESS in 6m 35s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9ea693abbbef43c697732926923b2f85 : SUCCESS in 6m 33s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2201abaa1ef34c07bf745811eca340b5 : SUCCESS in 13m 10s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a28140dbe61b47b3a1d4d6489ba41aad : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0acda866ad3e4d2eac4d3b11023afef0 : SUCCESS in 5m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2fd96885be63471ea128442d3fd5134e : SUCCESS in 6m 25s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6b96a151c2d948a7ad5e8c0bf316fc89 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/763efefd8c184a098e2c7e11e39080e9 : SUCCESS in 8m 34s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5dc58a254e4f4c8a978bcc1f19835ca5 : SUCCESS in 40m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c28996d512344f08baed3e82bf621ca2 : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a62470d3ff0e4c58a09ccdc01c0dcb68 : SUCCESS in 1h 43m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7fcac4f225f848aaa542b5d7ce23c571 : FAILURE in 1h 29m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3cc04d312014a74898ade0758aa90c8 : SUCCESS in 1h 44m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/b501b779c4924241995fd1df67da7b28 : FAILURE in 2h 25m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/8f98b7b158164ab6b67a24ded49dcc3d : SUCCESS in 2h 02m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/f93bdb5aaa2d47d8a662e4af9a996918 : SUCCESS in 37m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1cb8768c36ac4cd4a79627d20374b71f : SUCCESS in 57m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9c96cd8cb194d80b0cd5b8479fe7407 : SUCCESS in 1h 04m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dab2a32565c74b09b1d476771f34187f : SUCCESS in 24m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/602ffa96762d47f485ac283af51dc391 : SUCCESS in 1h 30m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/352a8890227049f893e8b629c53e620a : SUCCESS in 30m 43s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":26},{"id":"c5e20d3d2fb1ccf99b9e58287884fe44c2376fd8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-26 15:54:58.000000000","message":"Patch Set 26:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_09_2026_15_49_01\nLeave a comment with","accounts_in_message":[],"_revision_number":26},{"id":"048a5c2888e0b386e0afece7d31391c774da5ee5","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-28 22:46:32.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-26-fc : SUCCESS in 47m 59s\n- hitachi_ci_nvmetcp https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-26-nvmetcp : SUCCESS in 52m 18s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-26-iscsi : SUCCESS in 1h 35m 48s","accounts_in_message":[],"_revision_number":26},{"id":"2bc194f78a0031925d2fb5d6dcead0930d1814ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-29 07:59:53.000000000","message":"Patch Set 27: Published edit on patch set 26.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"b115637ba7eaed65f73771f8662784b960c3e915","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-09-29 08:00:09.000000000","message":"Patch Set 27:\n\n(4 comments)","accounts_in_message":[],"_revision_number":27},{"id":"2b1c37907219365819a13f154d1834b08af8a2a5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-29 08:16:42.000000000","message":"Patch Set 27:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/Cinder-CI/tree/gh-pages/865bb17b887a47cc95cb6c441ba84d87/","accounts_in_message":[],"_revision_number":27},{"id":"bbfab05b0c76b5023836611e9deb3aa4b95942b4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-29 08:42:34.000000000","message":"Patch Set 27:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/865bb17b887a47cc95cb6c441ba84d87\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81b0b33d0d16450ca7a6e62682c6fb9b : SUCCESS in 39m 21s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8bd38e8bb0f49d8bacab726ecfb8267 : SUCCESS in 40m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7ff8ec51d324dffbb837a346445b0e8 : SUCCESS in 42m 10s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2f138a3501c9466591002ad1e5236357 : SUCCESS in 30m 03s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cc7fb2bd476f4aaa82bb14e158f0e1a2 : FAILURE in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d47bc7135a01c1fd32c69922a1da09686ab84b3e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-29 08:53:45.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/5f442f9f92cd41f1819534856cffb8dc : SUCCESS in 49m 05s\n- cinder-storpool-tempest-iscsi-multipath build/4dc112ed25a14ca0a074d16b308325fe : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":27},{"id":"4209170c89acc19f4be498e2461c1a31bfb46c09","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-29 08:54:34.000000000","message":"Patch Set 27:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8cce8cc53633be716a927809fc41088b3468f22c .","accounts_in_message":[],"_revision_number":27},{"id":"81c6c7ff3496d85735f67be51070009191c1bbb1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-09-29 08:55:17.000000000","message":"Patch Set 27:\n\nHPE AlletraMP CI\n\nBuild succeeded.\n\n- iSCSI    : SUCCESS in 0h 47m 10s\n- FC       : SUCCESS in 0h 50m 5s\n- NVMe/TCP : SUCCESS in 0h 47m 5s\n\nLogs: https://github.com/hpe-storage/hpe_cinder_jenkins_logs/tree/master/logs/997469","accounts_in_message":[],"_revision_number":27},{"id":"b3b99d067f1319823e34873a28cab9862f0d210f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 09:31:35.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f41bf8bc0114b63816ee51bb6e0d259\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5058cccd8b604e51a20a7caed4e04bb5 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4197deff336947a7b44b60485c10cfce : TIMED_OUT in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"56d14a6ba10a8345099202201075c01b4703d75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 12:44:22.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbcca005420c4ead9a10d976b9ea9dfc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46f63aff10e2461298e5df9552b7afb1 : SUCCESS in 8m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64feed906a2543c1a94e5ef5e6ee7d37 : SUCCESS in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13ab3890dc064c6c88425205e3d16622 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d2e467f70349ae9685c05280739264 : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c71988e886514d9297368fb17929a8d1 : SUCCESS in 1h 04m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/546fdc1c79e64eed984c5c0bfd9afad7 : SUCCESS in 1h 06m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0851ec13c33e4c03807b220f92a854b7 : SUCCESS in 1h 30m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b41a037f6e62420ea65cd8fd1895d57f : SUCCESS in 39m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64760136d4f14e778b46a8cb44c1c58f : SUCCESS in 8m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4f1959f7ac141689d0934b307ca13cb : SUCCESS in 21m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4dd66037991c47ee87feaab6cd795a61 : SUCCESS in 6m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/31c8b682c1254fd7b1296cbd82c60845 : SUCCESS in 6m 52s (non-voting)\n- cinder-tox-py312-threading https://zuul.opendev.org/t/openstack/build/269ccfd6e8d84b0fa10c1884aeb3b524 : FAILURE in 11m 08s (non-voting)\n- cinder-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7ecbf8018365423aafa1ce5c34497e2f : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5cb9535d7f7e462fb716b4b4fc8e55af : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/336a5e9851f54ec2917f9cbb46ab7b38 : SUCCESS in 6m 40s\n- cinder-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a724c951f875476cad8b0b4b932e0c7b : FAILURE in 5m 34s (non-voting)\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/55702f7e32e64fcfabf85b91ec1b81d7 : SUCCESS in 9m 12s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1150e332898b43d4b9581f1975b52923 : SUCCESS in 37m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0da977fcac6a443fb9385166714e9456 : FAILURE in 5m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/178f6822cf444bb5890887247ba49548 : SUCCESS in 1h 33m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/463f8f5e813b4e57baf12f3a2377f986 : FAILURE in 2h 08m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a7553ee66b1448696b03661deb98e66 : SUCCESS in 1h 48m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-threading https://zuul.opendev.org/t/openstack/build/cc2b9b2600a042c5a030ebdb39b0d4b1 : FAILURE in 2h 40m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/de5524f6abd04258b7f5fd09fb0ec3b4 : SUCCESS in 2h 06m 46s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/e975e7a8319f4db0a054d911eb66e598 : SUCCESS in 25m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b85a042249d5417ea0f2f33b9063edec : SUCCESS in 1h 01m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/df5141baad8641b0994c065bc2618ec4 : SUCCESS in 1h 23m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9094e6d68b884f5398693b8eb0ef060e : SUCCESS in 44m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c8869b7281b48b69d68a493b1e2ca6f : SUCCESS in 1h 26m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce119ccc394b422b83bca9bbaedc6493 : SUCCESS in 59m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job cinder-tempest-plugin-lvm-lio-barbican-fips-py311.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":27},{"id":"644ea1400e7d9e75ab762deb0f48259d1cf0023c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-29 12:53:49.000000000","message":"Patch Set 27:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_09_2026_12_48_09\nLeave a comment with","accounts_in_message":[],"_revision_number":27},{"id":"b80d085d761ccc9834b06dd4d28e095dccf755bd","author":{"_account_id":38352,"name":"Hitachi CI","display_name":"Hitachi CI","email":"hitachi.ci.hbsd@gmail.com","username":"hitachi-ci-driver"},"date":"2026-09-29 21:29:26.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- hitachi_ci_fc https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-27-fc : SUCCESS in 1h 1m\n- hitachi_ci_nvmetcp https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-27-nvmetcp : SUCCESS in 1h 28s\n- hitachi_ci_iscsi https://github.com/hitachi-vantara/hbsd-tempest-result-files/tree/main/refs-changes-69-997469-27-iscsi : SUCCESS in 1h 9m 1s","accounts_in_message":[],"_revision_number":27},{"id":"f94b728f1d775b0e8729b1302f4b1c305fec8ad2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-30 03:14:52.000000000","message":"Patch Set 27:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5586/.","accounts_in_message":[],"_revision_number":27},{"id":"61e27e66ff8faf0bdb6e57824e49a4843b31eff3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-30 04:54:31.000000000","message":"Patch Set 27:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5763/.","accounts_in_message":[],"_revision_number":27},{"id":"1216093ad6694ffb5ed6edee48de59a85a04fb49","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-10-02 14:16:28.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"d6f55c938b6f58c53c5b62d19a2a5b7dd31bb42a","revisions":{"12ffcfa9b463d2630f404434c39fa2a4a3de61fd":{"kind":"REWORK","_number":1,"created":"2026-07-16 04:10:09.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/1"}}},"commit":{"parents":[{"commit":"8e5ea432eee90f1709395adf2d582da8986e7c42","subject":"[Hitachi] Fixes bug where reclaiming space does not occur automatically.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8e5ea432eee90f1709395adf2d582da8986e7c42"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12ffcfa9b463d2630f404434c39fa2a4a3de61fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/12ffcfa9b463d2630f404434c39fa2a4a3de61fd"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"205429e616c259074207f9f319d2e7185ab7a966":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-16 04:10:43.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/2"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:43.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/205429e616c259074207f9f319d2e7185ab7a966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/205429e616c259074207f9f319d2e7185ab7a966"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8e5ea432eee90f1709395adf2d582da8986e7c42","ours":"12ffcfa9b463d2630f404434c39fa2a4a3de61fd","theirs":"4bca3221ac40ce63c4681b5093312c44f1d490a1","merge_strategy":"recursive","contains_conflicts":false}},"7751c76004e71ffa92b487433f9cd30e6bb3b8a0":{"kind":"REWORK","_number":3,"created":"2026-07-16 04:45:50.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/3"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:44:33.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7751c76004e71ffa92b487433f9cd30e6bb3b8a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7751c76004e71ffa92b487433f9cd30e6bb3b8a0"}]},"branch":"refs/heads/master"},"426ba568f88e0c55210c0a6a995ed2936994660a":{"kind":"REWORK","_number":4,"created":"2026-07-17 05:07:12.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/4"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 05:07:03.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/426ba568f88e0c55210c0a6a995ed2936994660a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/426ba568f88e0c55210c0a6a995ed2936994660a"}]},"branch":"refs/heads/master"},"49a5bfac2ea26ce24fc724b3a2edabd9b3a8de54":{"kind":"REWORK","_number":5,"created":"2026-07-17 05:59:23.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/5"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 05:59:23.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a5bfac2ea26ce24fc724b3a2edabd9b3a8de54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/49a5bfac2ea26ce24fc724b3a2edabd9b3a8de54"}]},"branch":"refs/heads/master"},"bae3b5e6d0d239393453151777fb3804099d31c9":{"kind":"REWORK","_number":6,"created":"2026-07-17 08:58:32.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/6"}}},"commit":{"parents":[{"commit":"4bca3221ac40ce63c4681b5093312c44f1d490a1","subject":"Merge \"Update policy scope enforcement testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4bca3221ac40ce63c4681b5093312c44f1d490a1"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-17 08:58:32.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bae3b5e6d0d239393453151777fb3804099d31c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bae3b5e6d0d239393453151777fb3804099d31c9"}]},"branch":"refs/heads/master"},"2ac84bb2cef2fde87699ed3dcc65500085625791":{"kind":"REWORK","_number":7,"created":"2026-07-24 00:18:38.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/7"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-24 00:18:38.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ac84bb2cef2fde87699ed3dcc65500085625791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ac84bb2cef2fde87699ed3dcc65500085625791"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4bca3221ac40ce63c4681b5093312c44f1d490a1","ours":"bae3b5e6d0d239393453151777fb3804099d31c9","theirs":"fe0bd3aad8876c1635014f3797705fd36710a815","merge_strategy":"recursive","contains_conflicts":true}},"a7dd2241c9c17ac31b8c84a92137b812b8f23861":{"kind":"REWORK","_number":8,"created":"2026-07-24 00:20:19.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/8"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-24 00:20:18.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7dd2241c9c17ac31b8c84a92137b812b8f23861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7dd2241c9c17ac31b8c84a92137b812b8f23861"}]},"branch":"refs/heads/master"},"d1fb057a25703cd94bd52b49c6c440834c8ecef4":{"kind":"REWORK","_number":9,"created":"2026-07-24 04:57:39.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/9"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-24 04:57:39.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1fb057a25703cd94bd52b49c6c440834c8ecef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d1fb057a25703cd94bd52b49c6c440834c8ecef4"}]},"branch":"refs/heads/master"},"b14933803eaaae92f203c730f70788393438389a":{"kind":"REWORK","_number":10,"created":"2026-07-29 12:36:24.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/10"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-29 12:36:24.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b14933803eaaae92f203c730f70788393438389a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b14933803eaaae92f203c730f70788393438389a"}]},"branch":"refs/heads/master"},"048776b2c25c844cafdc9cb41c84e29a8adcf021":{"kind":"REWORK","_number":11,"created":"2026-07-29 23:29:13.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/11"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-29 23:29:07.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/048776b2c25c844cafdc9cb41c84e29a8adcf021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/048776b2c25c844cafdc9cb41c84e29a8adcf021"}]},"branch":"refs/heads/master"},"501c67308a3d7113707afd7df7ed66b349867652":{"kind":"REWORK","_number":12,"created":"2026-07-31 10:46:15.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/12"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-31 10:46:11.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/501c67308a3d7113707afd7df7ed66b349867652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/501c67308a3d7113707afd7df7ed66b349867652"}]},"branch":"refs/heads/master"},"0d889b6001d01650b72c45b0ef231eec646e31f0":{"kind":"REWORK","_number":13,"created":"2026-07-31 14:27:49.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/13"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-31 14:27:48.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d889b6001d01650b72c45b0ef231eec646e31f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0d889b6001d01650b72c45b0ef231eec646e31f0"}]},"branch":"refs/heads/master"},"373a3a60ea929d97eff48f3a3f3fcc3ea1452083":{"kind":"REWORK","_number":14,"created":"2026-08-26 15:21:46.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/14"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-26 15:20:47.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/373a3a60ea929d97eff48f3a3f3fcc3ea1452083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/373a3a60ea929d97eff48f3a3f3fcc3ea1452083"}]},"branch":"refs/heads/master"},"c4da8ba74aeb763354faa721fc422f976c8b20f6":{"kind":"REWORK","_number":15,"created":"2026-08-26 15:28:18.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/15"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-26 15:28:17.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4da8ba74aeb763354faa721fc422f976c8b20f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c4da8ba74aeb763354faa721fc422f976c8b20f6"}]},"branch":"refs/heads/master"},"3d5a760611398060781ecb6224b4bbe86d31001f":{"kind":"REWORK","_number":16,"created":"2026-08-26 21:47:51.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/16"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-26 21:47:51.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d5a760611398060781ecb6224b4bbe86d31001f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3d5a760611398060781ecb6224b4bbe86d31001f"}]},"branch":"refs/heads/master"},"62e66217a1034046497dbf2c8d4b00335ea7716e":{"kind":"REWORK","_number":17,"created":"2026-08-28 22:10:06.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/17"}}},"commit":{"parents":[{"commit":"fe0bd3aad8876c1635014f3797705fd36710a815","subject":"Merge \"Tests: Remove duplicate key in qos dict\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fe0bd3aad8876c1635014f3797705fd36710a815"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 22:10:06.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62e66217a1034046497dbf2c8d4b00335ea7716e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/62e66217a1034046497dbf2c8d4b00335ea7716e"}]},"branch":"refs/heads/master"},"432cd968a807b2e9420c0d2cbf06bac9c6ad23f3":{"kind":"REWORK","_number":18,"created":"2026-08-28 22:10:49.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/18"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 22:10:49.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/432cd968a807b2e9420c0d2cbf06bac9c6ad23f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/432cd968a807b2e9420c0d2cbf06bac9c6ad23f3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fe0bd3aad8876c1635014f3797705fd36710a815","ours":"62e66217a1034046497dbf2c8d4b00335ea7716e","theirs":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","merge_strategy":"recursive","contains_conflicts":true}},"25986d4d1c4168529bdc1726dfba2302a861235e":{"kind":"REWORK","_number":19,"created":"2026-08-28 22:18:40.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/19"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 22:18:35.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25986d4d1c4168529bdc1726dfba2302a861235e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25986d4d1c4168529bdc1726dfba2302a861235e"}]},"branch":"refs/heads/master"},"e63ba3609cbaed71f171eca90a88fc8e0b29d2ca":{"kind":"REWORK","_number":20,"created":"2026-08-28 22:36:45.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/20"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-08-28 22:36:45.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e63ba3609cbaed71f171eca90a88fc8e0b29d2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e63ba3609cbaed71f171eca90a88fc8e0b29d2ca"}]},"branch":"refs/heads/master"},"d71adbf0fa8132f8c5435fa846d1901568656a55":{"kind":"REWORK","_number":21,"created":"2026-09-17 23:28:13.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/21"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-17 23:28:06.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71adbf0fa8132f8c5435fa846d1901568656a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d71adbf0fa8132f8c5435fa846d1901568656a55"}]},"branch":"refs/heads/master"},"3f01c3c784303506f0770f30851371a4d5b16714":{"kind":"REWORK","_number":22,"created":"2026-09-18 03:11:17.000000000","uploader":{"_account_id":37940,"name":"Hao Chen","display_name":"Hao Chen","email":"hao.chen1@hitachivantara.com","username":"hvlcchao1"},"ref":"refs/changes/69/997469/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/22"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Hao Chen","email":"hao.chen1@hitachivantara.com","date":"2026-09-18 03:11:17.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f01c3c784303506f0770f30851371a4d5b16714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f01c3c784303506f0770f30851371a4d5b16714"}]},"branch":"refs/heads/master"},"b74b6e6d4859b42622fe0e5e593a295992bc668a":{"kind":"REWORK","_number":23,"created":"2026-09-26 09:36:54.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/23"}}},"commit":{"parents":[{"commit":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","subject":"Merge \"NetApp ONTAP REST: restore snapshot-dir fallback for CG clones\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dbb4e275c8547b9a714fc828b2a3752a77ec21fa"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-26 09:36:38.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication or\ncross-host multiattach.\n\nUpdated some out of date license and snapshot extension information\nin documentation.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74b6e6d4859b42622fe0e5e593a295992bc668a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b74b6e6d4859b42622fe0e5e593a295992bc668a"}]},"branch":"refs/heads/master"},"e1183104b39b7c9542a9edf7ad5b8b4c968ce676":{"kind":"REWORK","_number":24,"created":"2026-09-26 09:37:52.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/24"}}},"commit":{"parents":[{"commit":"4823af721c5855a56cf9fb6df3b496883e7a0258","subject":"Merge \"[Hitachi] Update version in python code.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4823af721c5855a56cf9fb6df3b496883e7a0258"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-26 09:37:52.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication or\ncross-host multiattach.\n\nUpdated some out of date license and snapshot extension information\nin documentation.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1183104b39b7c9542a9edf7ad5b8b4c968ce676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e1183104b39b7c9542a9edf7ad5b8b4c968ce676"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"dbb4e275c8547b9a714fc828b2a3752a77ec21fa","ours":"b74b6e6d4859b42622fe0e5e593a295992bc668a","theirs":"4823af721c5855a56cf9fb6df3b496883e7a0258","merge_strategy":"recursive","contains_conflicts":true}},"c23246b4766f2f813ef904ec433ef185c84dbe68":{"kind":"REWORK","_number":25,"created":"2026-09-26 09:38:26.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/25"}}},"commit":{"parents":[{"commit":"4823af721c5855a56cf9fb6df3b496883e7a0258","subject":"Merge \"[Hitachi] Update version in python code.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4823af721c5855a56cf9fb6df3b496883e7a0258"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-26 09:38:25.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication or\ncross-host multiattach.\n\nUpdated some out of date license and snapshot extension information\nin documentation.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c23246b4766f2f813ef904ec433ef185c84dbe68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c23246b4766f2f813ef904ec433ef185c84dbe68"}]},"branch":"refs/heads/master"},"5ffe3577817409df306cd4f0a6d1976493149495":{"kind":"REWORK","_number":26,"created":"2026-09-26 13:12:08.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/26"}}},"commit":{"parents":[{"commit":"4823af721c5855a56cf9fb6df3b496883e7a0258","subject":"Merge \"[Hitachi] Update version in python code.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4823af721c5855a56cf9fb6df3b496883e7a0258"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-26 13:12:07.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication or\ncross-host multiattach.\n\nUpdated some out of date license and snapshot extension information\nin documentation.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ffe3577817409df306cd4f0a6d1976493149495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ffe3577817409df306cd4f0a6d1976493149495"}]},"branch":"refs/heads/master"},"d6f55c938b6f58c53c5b62d19a2a5b7dd31bb42a":{"kind":"REWORK","_number":27,"created":"2026-09-29 07:59:53.000000000","uploader":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"ref":"refs/changes/69/997469/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/997469/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/997469/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/997469/27"}}},"commit":{"parents":[{"commit":"4823af721c5855a56cf9fb6df3b496883e7a0258","subject":"Merge \"[Hitachi] Update version in python code.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4823af721c5855a56cf9fb6df3b496883e7a0258"}]}],"author":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-07-16 04:10:09.000000000","tz":0},"committer":{"name":"Anthony Galica","email":"anthony.galica@hitachivantara.com","date":"2026-09-29 07:59:47.000000000","tz":0},"subject":"[Hitachi] Add support for NVMe-TCP.","message":"[Hitachi] Add support for NVMe-TCP.\n\nThe initial NVMe-TCP release does not support replication or\ncross-host multiattach.\n\nUpdated some out of date license and snapshot extension information\nin documentation.\n\nChange-Id: I8cce8cc53633be716a927809fc41088b3468f22c\nSigned-off-by: Anthony Galica \u003canthony.galica@hitachivantara.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f55c938b6f58c53c5b62d19a2a5b7dd31bb42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6f55c938b6f58c53c5b62d19a2a5b7dd31bb42a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
